)]}'
{"id":"openstack%2Fnova~318334","triplet_id":"openstack%2Fnova~master~I212cbb44f46d7cb36b5d8c74a79065d38fc526d8","project":"openstack/nova","branch":"master","topic":"bp/check-destination-on-migrations-newton","hashtags":[],"change_id":"I212cbb44f46d7cb36b5d8c74a79065d38fc526d8","subject":"API change for verifying the scheduler when live migrating","status":"MERGED","created":"2016-05-18 20:51:54.000000000","updated":"2016-06-06 14:46:34.000000000","submitted":"2016-06-06 14:46:32.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c30ae127bdc95ef3368bee61af56d281e69fdf65","_number":318334,"virtual_id_number":318334,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-06 14:46:32.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":1,"date":"2016-06-06 10:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-06 11:17:44.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-06 11:12:35.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-06 13:43:43.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-06 13:10:44.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-06 13:43:43.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-25 21:07:12.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2016-05-26 14:23:22.000000000","updated_by":{"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},"reviewer":{"_account_id":16495,"name":"Jean-Emile DARTOIS","email":"jean-emile.dartois@b-com.com","username":"JeD"},"state":"REVIEWER"},{"updated":"2016-05-26 16:31:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 13:43:25.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 14:42:30.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 16:35:39.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 17:06:46.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 18:36:01.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-02 20:21:38.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-03 11:22:32.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-03 19:56:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 10:20:11.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 11:12:35.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 11:17:44.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 13:10:44.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2016-06-06 13:43:43.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-06-06 14:46:32.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"a6e7adfe37774b5f3490ea0612089bcb435ca38b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-05-18 20:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc0cce5fa27fe5760f8805eb3b7a4a29c1c775ec","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-18 22:08:53.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/318334/1/15645/index.html : SUCCESS in 01 h 08 m 44 s","accounts_in_message":[],"_revision_number":1},{"id":"3d9db3cbe4d112fb423138684c666e3d3dd584c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-18 22:24:49.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/1/check/check-ibm-tempest-dsvm-full/28fb288/ : SUCCESS in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"719296411071bb4d88fd3356ed42eb58407d60ed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-18 22:45:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/1/check/gate-nova-docs/05352b5//doc/build/html/ : SUCCESS in 3m 47s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/1/check/gate-nova-pep8/a5979a0/ : SUCCESS in 5m 49s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/1/check/gate-nova-python27-db/d3d8988/ : FAILURE in 14m 38s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/1/check/gate-nova-python34-db/dfdcd44/ : FAILURE in 8m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-full/de410d4/ : SUCCESS in 1h 11m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-postgres-full/f4c4097/ : SUCCESS in 51m 36s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-neutron-full/dfeb6c1/ : SUCCESS in 1h 05m 55s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/1/check/gate-grenade-dsvm/800ed7c/ : SUCCESS in 42m 25s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/1/check/gate-nova-releasenotes/6bc9633//releasenotes/build/html/ : SUCCESS in 3m 40s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/1/check/gate-nova-api-ref/f2e07ac//api-ref/build/html/ : SUCCESS in 3m 42s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/1/check/gate-nova-tox-db-functional/4b1764c/ : SUCCESS in 12m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/1/check/gate-grenade-dsvm-multinode/30b688c/ : SUCCESS in 45m 05s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/afee2ef/ : SUCCESS in 42m 21s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-cells/cdcc8d1/ : SUCCESS in 33m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a241530/ : SUCCESS in 35m 32s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-multinode-full/ab58fc3/ : SUCCESS in 51m 12s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/2380d8a/ : SUCCESS in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b72099101c3d0594006ff88d68e7b82022a37480","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-19 00:55:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/1/check/dsvm-tempest-xen/1a94d6e : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d65e0c977a9195689240e6397aef23465299f54d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-19 01:12:32.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/318334/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d983bbbde61c66d586d5fee343809a85f3aff452","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 05:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/1/check/check-dsvm-tempest-vz7-exe-minimal/10c8188 : SUCCESS in 1h 25m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"469bb80fb446b34b2774950664da019556ab60f1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 10:58:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/1/check/tempest-dsvm-full-nfv/ce6ec9c : FAILURE in 1h 43m 21s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/1/check/tempest-dsvm-intel-nfv/38a0b00 : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":1},{"id":"70453ddb4a745aadb23cd5cd395dc62589db2d44","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-05-19 20:16:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a61f49f43924756e10a8b5c34d5d0168b7bce6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-19 22:04:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/2/check/gate-nova-docs/d7d7ca2//doc/build/html/ : SUCCESS in 6m 01s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/2/check/gate-nova-pep8/9ad47c5/ : SUCCESS in 7m 15s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/2/check/gate-nova-python27-db/d319f40/ : SUCCESS in 12m 07s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/2/check/gate-nova-python34-db/5abda60/ : SUCCESS in 12m 50s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-full/b3401cd/ : SUCCESS in 43m 46s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-postgres-full/33dcb78/ : SUCCESS in 35m 10s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-neutron-full/814adba/ : SUCCESS in 1h 06m 06s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/2/check/gate-grenade-dsvm/d67e940/ : SUCCESS in 25m 31s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/2/check/gate-nova-releasenotes/f6543c4//releasenotes/build/html/ : SUCCESS in 6m 29s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/2/check/gate-nova-api-ref/39e6548//api-ref/build/html/ : SUCCESS in 4m 21s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/2/check/gate-nova-tox-db-functional/6299145/ : SUCCESS in 14m 33s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/2/check/gate-grenade-dsvm-multinode/792dd8c/ : SUCCESS in 51m 22s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f48b119/ : SUCCESS in 37m 13s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-cells/25133d7/ : SUCCESS in 27m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/223cd7b/ : SUCCESS in 58m 06s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-multinode-full/0892d98/ : SUCCESS in 1h 09m 52s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/9f810ea/ : SUCCESS in 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22c44033b387925ce7513f60e415db037a01ae89","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-19 22:34:18.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/2/check/check-ibm-tempest-dsvm-full/e247782/ : SUCCESS in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1263d23dc05fb0b40af1e7af8d3e7b85f3b3534e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-19 22:40:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/2/check/dsvm-tempest-xen/877e668 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a7ae5ae51e868804da05e5d488d84c5abcfabe22","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-19 22:45:22.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/318334/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"1fe93ba27fee649882095b56bf4cd930af74e739","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-19 22:49:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/2/check/tempest-dsvm-full-nfv/f56f119 : FAILURE in 1h 37m 16s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/2/check/tempest-dsvm-intel-nfv/69db303 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1762b1704e968524fc696f6dcf23bf6fee871e4e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 05:09:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/2/check/check-dsvm-tempest-vz7-exe-minimal/3486654 : SUCCESS in 2h 02m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"0c2fc2b6fbed57f4330a1fb0627918dbd3be0a3f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-20 18:03:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/318334/2/4649 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a4bb3630f9d646bc25a1ba4a58e0b1d312ec5c2a","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2016-05-20 22:42:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)\n\nCouple of grammar nits, but LGTM","accounts_in_message":[],"_revision_number":2},{"id":"5ac2fcab18064e41efe734f52e8cbed5774135ad","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-22 15:18:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/34/318334/2/check-nova/Nova-ML2-Sriov/d71098f : SUCCESS in 47m 07s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/34/318334/2/check-nova/Nova-MACVTAP-ML2-Sriov/7ac9636 : SUCCESS in 46m 34s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"24334a40c8d3397d329ba34654b20ad860961732","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-23 03:44:15.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/ec2a4cc3309f410eb44028a3406e49de/index.html : SUCCESS in 15m 29s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e389948adb0645b38d6efe19a8cfc056/index.html : SUCCESS in 11m 26s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/55d1d9b72c7f412d9dab83a28d5096c3/index.html : SUCCESS in 14m 06s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/2/check/real-db-upgrade_nova_percona_user_001:th-percona/12f3049de6b543dab01e71a49de76f8c/index.html : SUCCESS in 13m 09s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd5b95b5b677247a01830a3083342a380515f09d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-05-24 10:34:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25ff68a113754dcbf7f25bab2302ee38807f5bed","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-24 10:55:04.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/e5e82e4f04cc49f2a7d90518f27cb06c/index.html : SUCCESS in 11m 49s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/067aa3cb2caf4061bb9bff30d29d3e2b/index.html : SUCCESS in 14m 19s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/c2413df952a546459e49559ddd9eefd6/index.html : SUCCESS in 17m 44s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/3/check/real-db-upgrade_nova_percona_user_001:th-percona/4e361a23db024194921f384b6acd59d0/index.html : SUCCESS in 11m 57s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"abc7fce24f8572350f1db490f90f58c17ffc9ead","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-24 11:07:35.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/318334/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"f4be528fed7c118e77a64179d64c5726f4ee8722","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-24 11:43:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/3/check/gate-nova-docs/595ec50//doc/build/html/ : SUCCESS in 4m 43s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/3/check/gate-nova-pep8/7e5882a/ : SUCCESS in 5m 46s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/3/check/gate-nova-python27-db/cbd0e28/ : SUCCESS in 6m 38s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/3/check/gate-nova-python34-db/b3659d4/ : SUCCESS in 11m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-full/49f6e58/ : SUCCESS in 46m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-postgres-full/5b2ba41/ : SUCCESS in 34m 57s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-neutron-full/ec1b592/ : FAILURE in 0s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/3/check/gate-grenade-dsvm/60c52eb/ : SUCCESS in 25m 35s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/3/check/gate-nova-releasenotes/6b53057//releasenotes/build/html/ : SUCCESS in 4m 34s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/3/check/gate-nova-api-ref/d66054a//api-ref/build/html/ : SUCCESS in 2m 21s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/3/check/gate-nova-tox-db-functional/7450dbc/ : SUCCESS in 12m 24s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/3/check/gate-grenade-dsvm-multinode/3ae927c/ : SUCCESS in 27m 55s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/f83bb5e/ : SUCCESS in 22m 14s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-cells/bf13acc/ : SUCCESS in 35m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a87b292/ : SUCCESS in 37m 13s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-multinode-full/392b963/ : FAILURE in 52m 50s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/b04adb4/ : SUCCESS in 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc63ddb57f2dc2fdb2b91b2250eb32a42eca3611","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-24 11:53:51.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/3/check/check-ibm-tempest-dsvm-full/c07cfc9/ : SUCCESS in 1h 18m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"262a07f9a770f85e0841d2aa2b4bd8a6fb09c540","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-24 12:00:37.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/318334/3/16192/index.html : SUCCESS in 01 h 16 m 02 s","accounts_in_message":[],"_revision_number":3},{"id":"54cdeff31e977ac712400973fea1a7081cff9481","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-24 12:05:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/3/check/dsvm-tempest-xen/b492dac : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":3},{"id":"bf4b5acc777fff4ec85c482630b13003cf875806","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 13:15:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/318334/3/ : SUCCESS in 2h 39m 22s","accounts_in_message":[],"_revision_number":3},{"id":"8d06707b9fe7f30aaebc560314df56b3dd0046a5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 15:03:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/3/check/check-dsvm-tempest-vz7-exe-minimal/89d9ee0 : SUCCESS in 2h 22m 42s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"f30754ec54c9aac96d7e4591a75e3b688c6775b8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-05-24 15:14:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/34/318334/3/check/dsvm-tempest-kvm/411155b : SUCCESS in 55m 35s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"452eead9eb24417bce8b632ca8811e3ef413a67a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-25 04:41:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/3/check/tempest-dsvm-full-nfv/56f1bb1 : FAILURE in 1h 45m 46s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/3/check/tempest-dsvm-intel-nfv/02333a5 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":3},{"id":"907ea4521c38392edf3cc15d0a7d9ee9435783a1","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-05-25 15:02:05.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nThis seems correct.","accounts_in_message":[],"_revision_number":3},{"id":"ec679df99dc03774bbaa449bb90138bd9a057ce3","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2016-05-25 21:07:12.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"307f144aa731f30c0f4300498a3814a03a3325b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-26 06:59:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild successful. \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/318334/3/4847 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"01f57214bcf0730940fc7dcf3f4fd09b79a99ab4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-05-26 11:58:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9fb2e6dc39f51f4fc1f56fc4dfe7546c6f28148","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-26 13:03:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/3/check/gate-nova-docs/3f8b225//doc/build/html/ : SUCCESS in 5m 20s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/3/check/gate-nova-pep8/1bbb15a/ : SUCCESS in 6m 16s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/3/check/gate-nova-python27-db/37449c6/ : SUCCESS in 10m 18s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/3/check/gate-nova-python34-db/292c918/ : SUCCESS in 9m 58s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-full/4848940/ : SUCCESS in 37m 03s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-postgres-full/92b8461/ : SUCCESS in 40m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-neutron-full/c2c6fed/ : SUCCESS in 1h 02m 06s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/3/check/gate-grenade-dsvm/a372622/ : SUCCESS in 24m 23s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/3/check/gate-nova-releasenotes/701d5f4//releasenotes/build/html/ : SUCCESS in 4m 50s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/3/check/gate-nova-api-ref/e086b65//api-ref/build/html/ : SUCCESS in 5m 35s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/3/check/gate-nova-tox-db-functional/ae21a20/ : SUCCESS in 10m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/3/check/gate-grenade-dsvm-multinode/6c57e43/ : SUCCESS in 45m 35s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/2917180/ : SUCCESS in 23m 19s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-cells/bd8fa17/ : SUCCESS in 30m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/b00f8a3/ : SUCCESS in 44m 48s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-multinode-full/4d68cc1/ : SUCCESS in 53m 04s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/0646f5b/ : SUCCESS in 57m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad408cbe2a8605ea1d91fa36940ea828f77580d9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-26 14:29:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/318334/3/ : SUCCESS in 2h 30m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6af2eff2768d9d26eb010cd4c7504b535d22b5d4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-05-26 16:31:21.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-34-318334-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"7e39556a1aaa27518c312a76582135876024c3be","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-26 17:36:31.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6bc23a45c72a79ad353b3232b41518e3e429e7bf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-05-27 08:20:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b34c0a5b142084cd306427422ac8b602e933adbd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-27 08:54:09.000000000","message":"Patch Set 4: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/318334/4/16648/index.html : FAILURE in 31 m 05 s","accounts_in_message":[],"_revision_number":4},{"id":"e78d72d67dcf3d02a29e7b4fc7dd6e3b1aab7537","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-27 09:04:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/4/check/dsvm-tempest-xen/f181cb0 : FAILURE in 31m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3c04ad65b412b893e4df4030bfa08bef88f6ba59","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-05-27 09:12:07.000000000","message":"Patch Set 4: Code-Review+2\n\nNasty rebase, but looks correct.","accounts_in_message":[],"_revision_number":4},{"id":"4571f4f04a36eb8a6bf06e82e3f59ec115894b7d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-27 09:16:27.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/318334/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"92ff3a70c16fab77708ebaa011dcf79459aaf66b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-27 09:29:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/4/check/gate-nova-docs/d992702//doc/build/html/ : SUCCESS in 2m 18s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/4/check/gate-nova-pep8/1e96222/ : SUCCESS in 5m 35s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/4/check/gate-nova-python27-db/8575fa5/ : SUCCESS in 8m 32s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/4/check/gate-nova-python34-db/5530bd1/ : SUCCESS in 8m 09s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-full/4257f8c/ : SUCCESS in 49m 24s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-postgres-full/dbebd17/ : SUCCESS in 39m 02s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-neutron-full/14d6e92/ : FAILURE in 58m 52s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/4/check/gate-grenade-dsvm/34a8df2/ : SUCCESS in 42m 11s\n- gate-nova-releasenotes http://logs.openstack.org/34/318334/4/check/gate-nova-releasenotes/c57ce40/ : FAILURE in 29m 58s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/4/check/gate-nova-api-ref/57e85ab//api-ref/build/html/ : SUCCESS in 3m 58s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/4/check/gate-nova-tox-db-functional/1213b86/ : SUCCESS in 10m 10s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/4/check/gate-grenade-dsvm-multinode/a601692/ : SUCCESS in 44m 43s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/ed899bd/ : FAILURE in 22m 48s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-cells/80bec06/ : SUCCESS in 42m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/bbf9844/ : SUCCESS in 1h 00m 18s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-multinode-full/2bcd78a/ : FAILURE in 1h 04m 06s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/312a647/ : SUCCESS in 55m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df193c452d3ab1a20d7c1b9b2ef66ce639eabe96","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-27 09:53:32.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/4/check/check-ibm-tempest-dsvm-full/78464e8/ : SUCCESS in 1h 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cd2b8a45ea35906e285fd0aebad224554d4e4073","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-05-27 11:21:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/34/318334/4/check/dsvm-tempest-kvm/f0fd193 : SUCCESS in 1h 05m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"d27ad3214ec710240cdc3a733a5a739986d18e8f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-05-27 11:21:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/318334/4/ : SUCCESS in 2h 18m 23s","accounts_in_message":[],"_revision_number":4},{"id":"907d2c7268ce82bbc069165e4efa46aaf7d8a9ea","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-27 16:53:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/4/check/check-dsvm-tempest-vz7-exe-minimal/654f43d : SUCCESS in 2h 29m 06s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"035e6b72160eb9de7473949eb800abab18f4d493","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-27 19:41:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/4/check/tempest-dsvm-full-nfv/56ab3f3 : FAILURE in 1h 28m 43s (non-voting)\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/4/check/tempest-dsvm-intel-nfv/75b5765 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":4},{"id":"192ea577739d7cd2b6a6040877f422e0eff4fc11","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2016-05-29 15:11:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"067a1d617aa628f32d779ece3ce1661fdb93574e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-29 16:09:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/4/check/gate-nova-docs/83e6e10//doc/build/html/ : SUCCESS in 3m 46s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/4/check/gate-nova-pep8/a681beb/ : SUCCESS in 5m 35s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/4/check/gate-nova-python27-db/d264a65/ : SUCCESS in 8m 19s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/4/check/gate-nova-python34-db/dc2c6e9/ : SUCCESS in 12m 04s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-full/062451f/ : SUCCESS in 43m 36s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-postgres-full/83ecc67/ : SUCCESS in 44m 13s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-neutron-full/5000c96/ : SUCCESS in 57m 48s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/4/check/gate-grenade-dsvm/3980614/ : SUCCESS in 37m 26s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/4/check/gate-nova-releasenotes/6d8a9bd//releasenotes/build/html/ : SUCCESS in 4m 18s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/4/check/gate-nova-api-ref/2a85a7c//api-ref/build/html/ : SUCCESS in 3m 54s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/4/check/gate-nova-tox-db-functional/d41841a/ : SUCCESS in 9m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/4/check/gate-grenade-dsvm-multinode/180ff1e/ : SUCCESS in 46m 45s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/0d3b090/ : FAILURE in 27m 37s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-cells/1a5c562/ : SUCCESS in 40m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/a363362/ : SUCCESS in 48m 13s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-multinode-full/1ccc8d0/ : SUCCESS in 57m 49s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/f1472ef/ : SUCCESS in 57m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"660953b806931ce136af4f09fb2a9079ab5ec3a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-01 11:16:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"212dc987908870ec01d06c0e3e9e4afda0ebe58c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-01 12:39:19.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"bbdf6dfb5f910b7e5a0307f4158033c8aa273660","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-06-01 14:18:28.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/318334/4/4972 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"819892d2cbcda17308d47aacea636f40ba9bd15f","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-01 16:53:14.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b78dbd31b4c54efb814f450ac41e61ea/index.html : SUCCESS in 18m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d5ecb6520bdb4c8785d249952e9dbdfd/index.html : SUCCESS in 22m 04s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/4656f8afcbf947a793c947021153cd33/index.html : SUCCESS in 21m 02s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/4/check/real-db-upgrade_nova_percona_user_001:th-percona/3d0bc859e9694e50a4d30f5d0fe9e7f6/index.html : SUCCESS in 16m 13s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":4},{"id":"86a550491ed8263f2c3120106af92d878abe7022","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-06-02 10:41:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f4a323699016bad09db6e5100234879243ad42da","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-02 10:42:18.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"00c8af9102798e5d27d324ede57aac84974e1a45","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-06-02 13:13:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"22383d496993a2be84f28761852ce177416a8f3b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-02 13:36:54.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/1bda298209884dcbb7a00bed1feaf2e9/index.html : SUCCESS in 20m 06s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/f38c5a99abc74adca62134d3a810dc1f/index.html : SUCCESS in 20m 39s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7528bf7406384a9b82f36abf20040869/index.html : SUCCESS in 19m 49s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/5/check/real-db-upgrade_nova_percona_user_001:th-percona/9dac526fa81a43b5a79ff77781b140e1/index.html : SUCCESS in 21m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":5},{"id":"2c7efff76c441450b8bf3cf2461940431df8a099","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-02 13:43:25.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/318334/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e9607359539379c2b18655d51990926d788064b7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-02 14:27:07.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/318334/5/17327/index.html : SUCCESS in 01 h 11 m 21 s","accounts_in_message":[],"_revision_number":5},{"id":"5ed5d4f82cabc9f87f3fbed68ae2a972dc28dec2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-02 14:33:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/5/check/dsvm-tempest-xen/0690df0 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":5},{"id":"2a2d0e3dde6b8b893e0b294a4e35b027fe4c1c59","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 14:42:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/5/check/tempest-dsvm-full-nfv/3c93ee5 : SUCCESS in 1h 27m 00s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/5/check/tempest-dsvm-intel-nfv/7ec7242 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":5},{"id":"72d2d6eae664dfdb5ea2a351314adda7a5bc6e05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-02 14:43:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/5/check/gate-nova-docs/8d27ec6//doc/build/html/ : SUCCESS in 3m 50s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/5/check/gate-nova-pep8/18c3353/ : SUCCESS in 4m 59s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/5/check/gate-nova-python27-db/6e3f4da/ : SUCCESS in 8m 35s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/5/check/gate-nova-python34-db/5a70385/ : SUCCESS in 11m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-full/8b4c615/ : SUCCESS in 47m 45s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-postgres-full/13f5694/ : SUCCESS in 48m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-neutron-full/c66bb93/ : SUCCESS in 1h 12m 48s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/5/check/gate-grenade-dsvm/cfaa416/ : SUCCESS in 41m 18s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/5/check/gate-nova-releasenotes/6789a3c//releasenotes/build/html/ : SUCCESS in 2m 43s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/5/check/gate-nova-api-ref/389646e//api-ref/build/html/ : SUCCESS in 2m 51s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/5/check/gate-nova-tox-db-functional/a6b5994/ : FAILURE in 9m 38s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/5/check/gate-grenade-dsvm-multinode/e1b4750/ : SUCCESS in 54m 19s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/b54dd58/ : SUCCESS in 50m 14s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-cells/66dafc6/ : SUCCESS in 35m 33s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/ad281bb/ : SUCCESS in 50m 21s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-multinode-full/1691fa8/ : SUCCESS in 1h 09m 54s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/39f4098/ : SUCCESS in 1h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"45343214a25e914eca3718196e3cbc81eaa6d0e9","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2016-06-02 14:43:57.000000000","message":"Patch Set 5: Code-Review-1\n\n@Sylvain, you forget update the api sample file\u0027s path to 2.30, like \"doc/api_samples/os-migrate-server/v2.29/live-migrate-server.json\"","accounts_in_message":[],"_revision_number":5},{"id":"a8afb5adaff6b9a495d5e32452857d890e56505d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 16:35:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/318334/5/ : SUCCESS in 3h 21m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3a6afe9dcf1ce2566b59b611228a321b628c5026","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-02 17:06:46.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/5/check/check-ibm-tempest-dsvm-full/c93b491/ : SUCCESS in 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b7f90f037df5b61c0edebc0af3f3b2931b58bd7b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 18:36:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/34/318334/5/check/dsvm-tempest-kvm/483bc40 : SUCCESS in 1h 13m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"923173bf7dd3e0791876d30ad99d874c321c1372","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-02 20:21:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/5/check/check-dsvm-tempest-vz7-exe-minimal/ed04cb8 : SUCCESS in 1h 21m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"5f75d0089f324ff7afc86ac209443733eddff0ee","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-06-02 20:40:01.000000000","message":"Patch Set 5: Workflow-1\n\nSnap...","accounts_in_message":[],"_revision_number":5},{"id":"714cfcaa7eba7b6047bf65632f9b6169ce418644","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-03 11:22:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/34/318334/5/check-nova/Nova-ML2-Sriov/17c8f6c : SUCCESS in 31m 26s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/34/318334/5/check-nova/Nova-MACVTAP-ML2-Sriov/b24de36 : SUCCESS in 36m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"bccee5766a1916b658b1d4360d41c23f13f7d6be","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-06-03 19:56:19.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n[http://file01.eng.vmware.com/ci-logs/dsvm-ext-nova/318334/5/104/ dsvm-ext-nova] : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"d28ea3929e340d32aaea4e85d236bbff8f06c85a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2016-06-06 09:55:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0efa82b780ead49e387aeb2d75997392e7f330d1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-06 10:20:11.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/726d0a90454c4555b5d38f8302b6694f/index.html : SUCCESS in 17m 16s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/0f7d9166b5cd4f75917956abed5f50a0/index.html : SUCCESS in 15m 32s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/34/318334/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/ac5e3620c7bf4b2086442addd29322a7/index.html : SUCCESS in 17m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/34/318334/6/check/real-db-upgrade_nova_percona_user_001:th-percona/53995d1d84894222b2a7adf5b38e63f5/index.html : SUCCESS in 23m 08s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":6},{"id":"0c4db45f479bb7d4b3b40a7a5ee817aab1fe20b1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-06 10:57:41.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/318334/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"46979ce6c1fc6690b30d9ec21ce2f0a0ec0f96ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-06 11:03:17.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/34/318334/6/check/check-ibm-tempest-dsvm-full/c05126a/ : SUCCESS in 1h 07m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9f088f46b1817464e00bc000611976bd70beb610","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 11:08:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/6/check/gate-nova-docs/08c8d90//doc/build/html/ : SUCCESS in 6m 01s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/6/check/gate-nova-pep8/bca5440/ : SUCCESS in 8m 35s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/6/check/gate-nova-python27-db/86fdf53/ : SUCCESS in 10m 54s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/6/check/gate-nova-python34-db/f9bfaf6/ : SUCCESS in 11m 54s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-full/56ba156/ : SUCCESS in 45m 23s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-postgres-full/69e9cc9/ : SUCCESS in 43m 59s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-neutron-full/5e48300/ : SUCCESS in 1h 00m 35s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/6/check/gate-grenade-dsvm/0c6d9c6/ : SUCCESS in 40m 36s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/6/check/gate-nova-releasenotes/02f1fae//releasenotes/build/html/ : SUCCESS in 5m 35s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/6/check/gate-nova-api-ref/4a587ea//api-ref/build/html/ : SUCCESS in 5m 40s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/6/check/gate-nova-tox-db-functional/7980d89/ : SUCCESS in 12m 58s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/6/check/gate-grenade-dsvm-multinode/a602447/ : SUCCESS in 43m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/70bbc16/ : SUCCESS in 34m 53s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-cells/1c4e490/ : SUCCESS in 36m 34s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph/f180285/ : SUCCESS in 48m 19s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-multinode-full/717ff26/ : SUCCESS in 57m 51s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/34/318334/6/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/604e951/ : SUCCESS in 1h 09m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"77a944a90d3c3b9e70ca17f0a111f59f02283a2d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-06 11:12:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/34/318334/6/check/dsvm-tempest-xen/7676c1f : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b4587d68cdf694529a83dac84566c665e3627b7a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-06 11:17:44.000000000","message":"Patch Set 6: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/318334/6/17640/index.html : SUCCESS in 01 h 19 m 07 s","accounts_in_message":[],"_revision_number":6},{"id":"5ae64dd802f88f4c7f0b5d21039c36e2e9f79958","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-06 11:23:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/34/318334/6/check/tempest-dsvm-full-nfv/9d95ac9 : SUCCESS in 1h 25m 22s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/34/318334/6/check/tempest-dsvm-intel-nfv/4f5f03b : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":6},{"id":"519f00e954f5d74bc07549bc28d6c3dcb28a7fb3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-06 12:05:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/34/318334/6/check/dsvm-tempest-kvm/51b10ea : SUCCESS in 1h 02m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"ffee8f80e5121058308b2887d5e7e7802c777725","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-06 12:46:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/318334/6/ : SUCCESS in 2h 50m 34s","accounts_in_message":[],"_revision_number":6},{"id":"c1955b1d23e0b489ae82f08e4b67e972230a5e32","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2016-06-06 13:10:44.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"82042390829dd8833111d8c4af314c11cd67fff0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-06-06 13:43:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/34/318334/6/check-nova/Nova-ML2-Sriov/fa5f5c8 : SUCCESS in 33m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/34/318334/6/check-nova/Nova-MACVTAP-ML2-Sriov/be065a7 : SUCCESS in 34m 09s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"a97f28fd1e4115299775d92f5e66ee0f8f18ab1d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-06-06 13:43:43.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLooks good now, thank you!","accounts_in_message":[],"_revision_number":6},{"id":"11d8cebd57d8dc4be75e0123b31c78a15fc73315","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 13:47:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"3a255481b58c1ddfb8868ad2973c5a1036ec6dab","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-06 14:19:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/34/318334/6/check/check-dsvm-tempest-vz7-exe-minimal/ce79e5d : SUCCESS in 2h 24m 01s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"20fe55127121b664804acfacc39f4d3fb6ba7b24","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-06 14:46:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/34/318334/6/gate/gate-nova-docs/19f37a1//doc/build/html/ : SUCCESS in 6m 32s\n- gate-nova-pep8 http://logs.openstack.org/34/318334/6/gate/gate-nova-pep8/aa2dd0b/ : SUCCESS in 6m 37s\n- gate-nova-python27-db http://logs.openstack.org/34/318334/6/gate/gate-nova-python27-db/ccac05e/ : SUCCESS in 13m 06s\n- gate-nova-python34-db http://logs.openstack.org/34/318334/6/gate/gate-nova-python34-db/562bce6/ : SUCCESS in 8m 39s\n- gate-tempest-dsvm-full http://logs.openstack.org/34/318334/6/gate/gate-tempest-dsvm-full/c09a15d/ : SUCCESS in 36m 42s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/318334/6/gate/gate-tempest-dsvm-postgres-full/f1aaf5f/ : SUCCESS in 38m 07s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/34/318334/6/gate/gate-tempest-dsvm-neutron-full/3595373/ : SUCCESS in 55m 24s\n- gate-grenade-dsvm http://logs.openstack.org/34/318334/6/gate/gate-grenade-dsvm/be28fd5/ : SUCCESS in 25m 21s\n- gate-nova-releasenotes http://docs-draft.openstack.org/34/318334/6/gate/gate-nova-releasenotes/8f7d0eb//releasenotes/build/html/ : SUCCESS in 3m 48s\n- gate-nova-api-ref http://docs-draft.openstack.org/34/318334/6/gate/gate-nova-api-ref/836d742//api-ref/build/html/ : SUCCESS in 3m 48s\n- gate-nova-tox-db-functional http://logs.openstack.org/34/318334/6/gate/gate-nova-tox-db-functional/440def4/ : SUCCESS in 12m 56s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/34/318334/6/gate/gate-grenade-dsvm-multinode/8a92d18/ : SUCCESS in 46m 20s\n- gate-tempest-dsvm-cells http://logs.openstack.org/34/318334/6/gate/gate-tempest-dsvm-cells/3839940/ : SUCCESS in 28m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/34/318334/6/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/1c86551/ : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":6},{"id":"85139b5d7b1f9f65b42647256d73cac0335c19ca","date":"2016-06-06 14:46:34.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7aa2285e724345717a3f333adc13660d7b97dfcd","revisions":{"0269d8385f7896d75bdffd02bde4013627124c66":{"kind":"REWORK","_number":1,"created":"2016-05-18 20:51:54.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/1"}}},"commit":{"parents":[{"commit":"fe8937273fb14abf22ba5ea638f0fa7e866aaa84","subject":"WIP: API change for verifying the scheduler when evacuating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe8937273fb14abf22ba5ea638f0fa7e866aaa84"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:44:11.000000000","tz":120},"subject":"WIP: API change for verifying the scheduler when live migrating","message":"WIP: API change for verifying the scheduler when live migrating\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0269d8385f7896d75bdffd02bde4013627124c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0269d8385f7896d75bdffd02bde4013627124c66"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fe8937273fb14abf22ba5ea638f0fa7e866aaa84","is_merged_in_target_branch":false,"change_id":"I9ecbe3d481bf17b12072511da4bb106ff1b6404e","change_number":315572,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8c00611d94b6d61f53637b3c41b1d6e36b0b22da":{"kind":"REWORK","_number":2,"created":"2016-05-19 20:16:22.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/2"}}},"commit":{"parents":[{"commit":"fc3401c497f552dfbe25c74d7f7211741d87d2d1","subject":"API change for verifying the scheduler when evacuating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc3401c497f552dfbe25c74d7f7211741d87d2d1"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-19 20:11:46.000000000","tz":120},"subject":"API change for verifying the scheduler when live migrating","message":"API change for verifying the scheduler when live migrating\n\nAfter modifying the evacuate action, we now add a new microversion\nchange for modifying the live-migrate call so that the scheduler is\ncalled when the admin user provides an hostname unless the force\nfield is provided.\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c00611d94b6d61f53637b3c41b1d6e36b0b22da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c00611d94b6d61f53637b3c41b1d6e36b0b22da"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc3401c497f552dfbe25c74d7f7211741d87d2d1","is_merged_in_target_branch":false,"change_id":"I9ecbe3d481bf17b12072511da4bb106ff1b6404e","change_number":315572,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8a7d90eec0afae7ddbeac87654ec56159e99321c":{"kind":"REWORK","_number":3,"created":"2016-05-24 10:34:32.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/3"}}},"commit":{"parents":[{"commit":"353bcb9bc4114e203bb0df3fdf835cba98873d01","subject":"API change for verifying the scheduler when evacuating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/353bcb9bc4114e203bb0df3fdf835cba98873d01"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-24 10:33:44.000000000","tz":120},"subject":"API change for verifying the scheduler when live migrating","message":"API change for verifying the scheduler when live migrating\n\nAfter modifying the evacuate action, we now add a new microversion\nchange for modifying the live-migrate call so that the scheduler is\ncalled when the admin user provides an hostname unless the force\nfield is provided.\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a7d90eec0afae7ddbeac87654ec56159e99321c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a7d90eec0afae7ddbeac87654ec56159e99321c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"353bcb9bc4114e203bb0df3fdf835cba98873d01","is_merged_in_target_branch":false,"change_id":"I9ecbe3d481bf17b12072511da4bb106ff1b6404e","change_number":315572,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dd0b905f9967c0b69b2e32dda2f8daa988bdfc86":{"kind":"REWORK","_number":4,"created":"2016-05-27 08:20:31.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/4"}}},"commit":{"parents":[{"commit":"654af4b2d48eaccc9e6f0b27b27cd5f1c10438c8","subject":"API change for verifying the scheduler when evacuating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/654af4b2d48eaccc9e6f0b27b27cd5f1c10438c8"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-27 08:19:53.000000000","tz":120},"subject":"API change for verifying the scheduler when live migrating","message":"API change for verifying the scheduler when live migrating\n\nAfter modifying the evacuate action, we now add a new microversion\nchange for modifying the live-migrate call so that the scheduler is\ncalled when the admin user provides an hostname unless the force\nfield is provided.\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd0b905f9967c0b69b2e32dda2f8daa988bdfc86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd0b905f9967c0b69b2e32dda2f8daa988bdfc86"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"654af4b2d48eaccc9e6f0b27b27cd5f1c10438c8","is_merged_in_target_branch":false,"change_id":"I9ecbe3d481bf17b12072511da4bb106ff1b6404e","change_number":315572,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9e64e0b037f2aa063507cb5049f656b8921ad709":{"kind":"REWORK","_number":5,"created":"2016-06-02 13:13:23.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/5"}}},"commit":{"parents":[{"commit":"86706785ff251b841dff3590dc60f6b4834d7b7e","subject":"API change for verifying the scheduler when evacuating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/86706785ff251b841dff3590dc60f6b4834d7b7e"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-06-02 13:11:03.000000000","tz":120},"subject":"API change for verifying the scheduler when live migrating","message":"API change for verifying the scheduler when live migrating\n\nAfter modifying the evacuate action, we now add a new microversion\nchange for modifying the live-migrate call so that the scheduler is\ncalled when the admin user provides an hostname unless the force\nfield is provided.\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9e64e0b037f2aa063507cb5049f656b8921ad709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9e64e0b037f2aa063507cb5049f656b8921ad709"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"86706785ff251b841dff3590dc60f6b4834d7b7e","is_merged_in_target_branch":true,"change_id":"I9ecbe3d481bf17b12072511da4bb106ff1b6404e","change_number":315572,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7aa2285e724345717a3f333adc13660d7b97dfcd":{"kind":"REWORK","_number":6,"created":"2016-06-06 09:55:17.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/34/318334/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/34/318334/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/34/318334/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/34/318334/6"}}},"commit":{"parents":[{"commit":"545d8d8666389f33601b0b003dec844004694919","subject":"Merge \"Fall back to flat config drive if not found in rbd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/545d8d8666389f33601b0b003dec844004694919"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-05-18 20:00:23.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2016-06-06 09:44:06.000000000","tz":120},"subject":"API change for verifying the scheduler when live migrating","message":"API change for verifying the scheduler when live migrating\n\nAfter modifying the evacuate action, we now add a new microversion\nchange for modifying the live-migrate call so that the scheduler is\ncalled when the admin user provides an hostname unless the force\nfield is provided.\n\nAPIImpact\n\nImplements: blueprint check-destination-on-migrations-newton\n\nChange-Id: I212cbb44f46d7cb36b5d8c74a79065d38fc526d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7aa2285e724345717a3f333adc13660d7b97dfcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7aa2285e724345717a3f333adc13660d7b97dfcd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"545d8d8666389f33601b0b003dec844004694919","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
