)]}'
{"id":"openstack%2Fnova~319379","triplet_id":"openstack%2Fnova~master~I8742071b55f018f864f5a382de20075a5b444a79","project":"openstack/nova","branch":"master","topic":"bp/cells-scheduling-interaction","hashtags":[],"change_id":"I8742071b55f018f864f5a382de20075a5b444a79","subject":"Move instance creation to conductor","status":"MERGED","created":"2016-05-20 17:17:50.000000000","updated":"2017-03-18 01:37:30.000000000","submitted":"2017-01-27 08:47:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"31ec35e49e052cbc3a0849a96f6ad70b36468219","_number":319379,"virtual_id_number":319379,"owner":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-01-27 08:47:06.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":-1,"date":"2017-01-26 21:08:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":2,"date":"2017-01-26 19:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2017-01-26 19:56:14.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2017-01-26 19:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":1,"date":"2017-01-26 19:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-23 08:08:29.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 10:25:17.000000000","updated_by":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"reviewer":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"state":"REVIEWER"},{"updated":"2017-01-05 07:34:42.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-05 07:54:26.000000000","updated_by":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"reviewer":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"state":"REVIEWER"},{"updated":"2017-01-17 08:42:12.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-01-20 14:43:24.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2017-01-25 22:31:53.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 07:21:53.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 07:50:28.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 07:55:08.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 08:25:39.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 08:27:09.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2017-01-26 08:34:39.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 08:36:44.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 09:47:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 10:08:44.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-01-26 11:07:54.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-26 11:20:56.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2017-01-26 14:35:52.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2017-01-26 19:57:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2017-01-26 21:08:20.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-27 08:47:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-02-08 16:22:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6da148a8d876b157340a473f6c8bb9b6a3935b94","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-05-20 17:17:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"221361dac05d84cad1d07d120218ae7700e6a8da","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-05-20 18:18:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/1/check/dsvm-tempest-xen/f0ef74c : FAILURE in 45m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c82cf0d8c52c1aa4013ee943a643d07fc6653ffe","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-05-20 18:21:54.000000000","message":"Patch Set 1: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/1/15902/index.html : FAILURE in 58 m 22 s","accounts_in_message":[],"_revision_number":1},{"id":"32ea3a013bfe6dfc397a907e84c42a50229b3d92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-05-20 19:01:40.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/1/check/check-ibm-tempest-dsvm-full/417f9d6/ : FAILURE in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d5c70c511578eac0f2819f6f7e7e3662a1e1ce0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-20 19:10:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/79/319379/1/check/gate-nova-docs/bc1d004//doc/build/html/ : SUCCESS in 6m 07s\n- gate-nova-pep8 http://logs.openstack.org/79/319379/1/check/gate-nova-pep8/daed54c/ : FAILURE in 6m 18s\n- gate-nova-python27-db http://logs.openstack.org/79/319379/1/check/gate-nova-python27-db/f27350e/ : FAILURE in 10m 41s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/1/check/gate-nova-python34-db/a3cb4fa/ : FAILURE in 11m 43s\n- gate-tempest-dsvm-full http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-full/c0a2e20/ : FAILURE in 27m 20s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-postgres-full/749fabc/ : FAILURE in 26m 44s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-neutron-full/3113873/ : FAILURE in 38m 51s\n- gate-grenade-dsvm http://logs.openstack.org/79/319379/1/check/gate-grenade-dsvm/4e7f22f/ : FAILURE in 40m 49s\n- gate-nova-tox-db-functional http://logs.openstack.org/79/319379/1/check/gate-nova-tox-db-functional/5db974b/ : FAILURE in 8m 34s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/1/check/gate-grenade-dsvm-multinode/fa7646a/ : FAILURE in 46m 32s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6b8134c/ : FAILURE in 27m 09s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-cells/4fda39b/ : FAILURE in 28m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/00951d7/ : FAILURE in 30m 55s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-multinode-full/3832b3f/ : FAILURE in 58m 25s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/79/319379/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/3746c89/ : FAILURE in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"823c7bd64f643c374d9ab94f33759245b0c9c658","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-05-20 19:21:14.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/319379/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2645c97a7c8bb81d91ec1a9e0cd4db605de67c97","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 22:36:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/79/319379/1/check/tempest-dsvm-full-nfv/c0781cc : FAILURE in 48m 06s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/79/319379/1/check/tempest-dsvm-intel-nfv/aae0cab : FAILURE in 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"937b37854eea2d1634876ecd622220e369b684e8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-05-20 23:25:27.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/1/check/check-dsvm-tempest-vz7-exe-minimal/18e85a6 : FAILURE in 1h 02m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"d20dd666034df63d1912f0fe13f2874f093a8e7f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-05-21 17:42:29.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails  \n\n* nova-vc6-dsvm-tempest http://208.91.1.172/logs/nova-vc6-dsvm-tempest/319379/1/4689 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2770d2d5563376f7dccaf2b7edf47505b0212781","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-05-23 08:08:29.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/79/319379/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/f53c343c563d4d1e9308ff39080aa29c/index.html : SUCCESS in 13m 43s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/79/319379/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/e181e117987c4dccb44a3fd176eaa640/index.html : SUCCESS in 10m 58s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/79/319379/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/6b5d8d0d0bca4243843aaf47bd50352b/index.html : SUCCESS in 16m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/79/319379/1/check/real-db-upgrade_nova_percona_user_001:th-percona/7b345f92c5e74be0b408f50710e773c3/index.html : SUCCESS in 12m 02s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd901a53dea1396b2c9ebbcfe20ba3074b0c2890","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-05-23 20:19:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://144.76.193.39/ci-artifacts/79/319379/1/check-nova/Nova-ML2-Sriov/a32a9e1 : FAILURE in 36m 37s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://144.76.193.39/ci-artifacts/79/319379/1/check-nova/Nova-MACVTAP-ML2-Sriov/fdf25cc : FAILURE in 33m 10s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"7cd213deb135d4352a4969f4926ae8abca142c7d","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2016-07-26 20:16:17.000000000","message":"Abandoned\n\nThis code hasn\u0027t been updated in a long time, and is in merge conflict. I am going to abandon this review, but feel free to restore it if you\u0027re still working on this.","accounts_in_message":[],"_revision_number":1},{"id":"fe8d0c78498d9bf4e85d7081401de1cc53307bea","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-08-30 16:02:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"813a769f8c858eacc5e9096eb1661f884cb20d7a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-08-30 16:03:06.000000000","message":"Patch Set 1:\n\nMerge failed. To try again use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":1},{"id":"24d5f5b6b9d935d5d1e43d1540bc10a450338614","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-08-30 16:04:16.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9a0a568d2973fe609a6b71b4d09b4987d192aa0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 16:08:57.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"066acaab324b98d927d9e833156a58fd44c61b0b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-08-30 16:18:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"9bb6e62ade24f324ea325d73fd59abaa57756245","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-08-30 18:04:26.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6a0add06e3c0f4eb10efab20010264570b539a38","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-08-30 19:00:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5701012e9a1a135e96a0bdd0705249004658f90","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-30 19:42:20.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/319379/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1b7289b00e02972955f236c69682adf1fe6d4e7a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-08-30 19:44:17.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/2/check/check-ibm-tempest-dsvm-full/3f9e2b9/ : FAILURE in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"921e589f55abbc0bd561cf0a5aa9153df87041a3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-08-30 20:05:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/2/check/dsvm-tempest-xen/232b152 : FAILURE in 42m 03s","accounts_in_message":[],"_revision_number":2},{"id":"43be9e57119893ed089849cbc3b3ea1392039c85","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 20:18:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/2/check/gate-nova-docs-ubuntu-xenial/d1cf50a//doc/build/html/ : SUCCESS in 2m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/2/check/gate-nova-pep8-ubuntu-xenial/6857805/ : SUCCESS in 7m 23s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/2/check/gate-nova-python27-db-ubuntu-xenial/2bfdc03/ : FAILURE in 12m 34s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/2/check/gate-nova-python34-db/1da19e8/ : FAILURE in 9m 34s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/2/check/gate-nova-python35-db-nv/f89e5c1/ : FAILURE in 11m 22s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/2/check/gate-grenade-dsvm-neutron-ubuntu-trusty/ae5a6dc/ : FAILURE in 49m 38s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/91b132f/ : FAILURE in 49m 00s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-full-ubuntu-xenial/3a2d7b5/ : FAILURE in 35m 11s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/2/check/gate-grenade-dsvm-ubuntu-trusty/ce1cd06/ : FAILURE in 40m 01s\n- gate-nova-tox-db-functional http://logs.openstack.org/79/319379/2/check/gate-nova-tox-db-functional/e922449/ : FAILURE in 11m 38s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/2/check/gate-grenade-dsvm-multinode/63ec291/ : FAILURE in 39m 33s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/cf6c368/ : FAILURE in 35m 08s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-cells/e4f105a/ : FAILURE in 35m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/e52f77d/ : FAILURE in 36m 43s\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-neutron-multinode-full/b4047bc/ : FAILURE in 1h 03m 19s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/79/319379/2/check/gate-tempest-dsvm-multinode-live-migration/70b4b27/ : FAILURE in 27m 19s","accounts_in_message":[],"_revision_number":2},{"id":"632c0c0365c148bbac3992c05f2c31228a1504f7","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-08-30 20:49:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65f3104f59cadd7e7adf06391edd67653061a8da","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-30 21:25:38.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/319379/3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4f460e734b2e9da485ce4e0dadc35c86287aa5cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-08-30 21:31:24.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/3/check/check-ibm-tempest-dsvm-full/a2355b2/ : FAILURE in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"109d1221d39d8389d99c8e5dd8516b736b02fe45","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-08-30 21:42:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/3/check/dsvm-tempest-xen/e5d14eb : FAILURE in 52m 23s","accounts_in_message":[],"_revision_number":3},{"id":"62c181728f3320b85a43535b544b8479c61354c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 21:44:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/3/check/gate-nova-docs-ubuntu-xenial/7661576//doc/build/html/ : SUCCESS in 4m 25s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/3/check/gate-nova-pep8-ubuntu-xenial/d7b325e/ : SUCCESS in 5m 58s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/3/check/gate-nova-python27-db-ubuntu-xenial/343a8f7/ : FAILURE in 11m 22s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/3/check/gate-nova-python34-db/f6d7451/ : FAILURE in 9m 19s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/3/check/gate-nova-python35-db-nv/8501815/ : FAILURE in 12m 12s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/3/check/gate-grenade-dsvm-neutron-ubuntu-trusty/46e2049/ : FAILURE in 34m 15s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8dc8d8a/ : FAILURE in 46m 47s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-full-ubuntu-xenial/87831e5/ : FAILURE in 25m 24s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/3/check/gate-grenade-dsvm-ubuntu-trusty/cc9629d/ : FAILURE in 24m 46s\n- gate-nova-tox-db-functional http://logs.openstack.org/79/319379/3/check/gate-nova-tox-db-functional/7e2f9d5/ : FAILURE in 11m 56s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/3/check/gate-grenade-dsvm-multinode/5893cea/ : FAILURE in 27m 03s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/ccfadfe/ : FAILURE in 29m 21s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-cells/ce0ce46/ : FAILURE in 24m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/35443c7/ : FAILURE in 29m 46s\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-neutron-multinode-full/a8615fc/ : FAILURE in 49m 28s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/79/319379/3/check/gate-tempest-dsvm-multinode-live-migration/0be4541/ : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":3},{"id":"3bea0d33c6ebd9a60ed3c1e7280386d08d24bef0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-08-30 22:44:47.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/3/ : FAILURE in 1h 54m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3e87fed3d98e17db4875dd52473d2b2a3adc0c0a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-08-30 23:25:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/3/check/dsvm-tempest-nova-network/f294eab : SUCCESS in 1h 15m 32s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/3/check/dsvm-tempest-neutron-network/cc2467d : SUCCESS in 2h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e50411b170776cc2c483f39798951b38fc398ab1","author":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"date":"2016-08-31 05:48:44.000000000","message":"Patch Set 3:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":3},{"id":"831106d1506b45b88aff993b7ec745da2a03ab99","author":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"date":"2016-08-31 05:49:28.000000000","message":"Patch Set 2:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":2},{"id":"85f48a43bc48c2fc1bb49ea0723bab1ae1658bb8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-08-31 07:58:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/3/check/dsvm-tempest-kvm/60e0111 : FAILURE in 1h 03m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"358b31cbe99003962b7f8df114aa9b3d3866552a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-31 17:39:26.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"a5c4cf966f1d9a44872c8eb1c1ef68eff237a40f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-31 17:39:33.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"296a30a69a21a4801b62aab8bfbf96c0e19dc8e5","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-09-01 15:03:49.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8c6625eeaa69954c587d578ae662a9af71b9acb9","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-09-01 15:12:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5b397ab08dfa2bfaa3a60f6a235ebd0aa7c924d2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-01 15:27:22.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/319379/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3fb1ed1e9cf9c8a14d75340b5b500f840b25b9a2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-01 15:32:56.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/319379/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"ae048d83e7dfd2db3e87990b36f32fb154e8a68f","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-01 15:44:09.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7fcaeb0b16fcf80b44e2e9a965e325f6147dd347","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-01 15:57:50.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/5/check/check-ibm-tempest-dsvm-full/0a2c7df/ : FAILURE in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"18698253286fa786f662ac04f020c145262c3291","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-01 16:18:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/5/check/dsvm-tempest-xen/dda6ebe : FAILURE in 1h 02m 26s","accounts_in_message":[],"_revision_number":5},{"id":"ca0f5080c908cba1b15ba64557a717225bd70034","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-01 16:32:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/5/check/gate-nova-docs-ubuntu-xenial/489cef9//doc/build/html/ : SUCCESS in 5m 00s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/5/check/gate-nova-pep8-ubuntu-xenial/8a68762/ : SUCCESS in 7m 38s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/5/check/gate-nova-python27-db-ubuntu-xenial/59417d9/ : FAILURE in 13m 20s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/5/check/gate-nova-python34-db/1e03e4c/ : FAILURE in 10m 41s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/5/check/gate-nova-python35-db-nv/271a98e/ : FAILURE in 12m 26s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/5/check/gate-grenade-dsvm-neutron-ubuntu-trusty/0400db8/ : FAILURE in 31m 03s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/b7418ca/ : FAILURE in 41m 54s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-full-ubuntu-xenial/91fd8ee/ : FAILURE in 25m 06s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/5/check/gate-grenade-dsvm-ubuntu-trusty/f361a5b/ : FAILURE in 23m 16s\n- gate-nova-tox-db-functional http://logs.openstack.org/79/319379/5/check/gate-nova-tox-db-functional/44702b5/ : FAILURE in 12m 23s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/5/check/gate-grenade-dsvm-multinode/2a12a80/ : FAILURE in 41m 39s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/12c9002/ : FAILURE in 23m 42s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-cells/e008380/ : FAILURE in 26m 43s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/c57fa87/ : FAILURE in 29m 48s\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-neutron-multinode-full/5e05177/ : FAILURE in 1h 05m 04s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/c3e70f7/ : FAILURE in 40m 40s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/79/319379/5/check/gate-tempest-dsvm-multinode-live-migration/fdc9f87/ : FAILURE in 26m 34s","accounts_in_message":[],"_revision_number":5},{"id":"2b244bb757ed9d5b02381e16685ab646c535c733","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-01 17:01:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/5/check-nova/Nova-ML2-Sriov/903f395 : FAILURE in 26m 33s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/5/check-nova/Nova-MACVTAP-ML2-Sriov/6d0cc9e : FAILURE in 24m 43s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"7b97108a2138baaacb0dc88a92b4b60f349314d6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-01 17:56:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/5/check/dsvm-tempest-nova-network/b99fb79 : SUCCESS in 1h 37m 46s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/5/check/dsvm-tempest-neutron-network/d67202c : SUCCESS in 1h 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1384215ebc6c9ae4c1443fe1e642fc3563af5808","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-01 19:23:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-full-nfv/3cb0662 : FAILURE in 1h 07m 12s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-intel-nfv/a817c5e : FAILURE in 45m 52s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-ovsdpdk-nfv-networking/f2f8676 : FAILURE in 58m 01s\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-full-nfv-xenial/67cd4c5 : FAILURE in 1h 02m 30s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-intel-nfv-xenial/9050550 : FAILURE in 46m 39s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ad1f00d : FAILURE in 59m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a37a3b9e7420558225fe81842604cf5e8a1d9a6e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-03 10:05:55.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/5/557/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c311a202b2383519aafd98bb3ffc0d49cfa9d6e1","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-09-06 17:02:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8a2596a479fd2973952b298dceec51341bbd623c","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-06 17:58:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1bb01b8140b699e5cdd4bcba86aacf819765f793","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-06 18:18:47.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/319379/6 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"dad006f810d52009aeeea25a5dc9c87e2bc96f07","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-06 18:43:04.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/6/check/check-ibm-tempest-dsvm-full/734a898/ : FAILURE in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"10ad4557f13233792842133d0e43e0548cdfc315","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-06 18:57:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/6/check/gate-nova-docs-ubuntu-xenial/2f718d9//doc/build/html/ : SUCCESS in 2m 20s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-nova-pep8-ubuntu-xenial/d082ee9/ : SUCCESS in 4m 46s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-nova-python27-db-ubuntu-xenial/212c81e/ : FAILURE in 15m 46s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/6/check/gate-nova-python34-db/917a9de/ : FAILURE in 11m 32s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/6/check/gate-nova-python35-db-nv/ec7727b/ : FAILURE in 11m 48s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/6/check/gate-grenade-dsvm-neutron-ubuntu-trusty/1205df6/ : FAILURE in 38m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a4965be/ : FAILURE in 39m 13s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-full-ubuntu-xenial/6ad56ff/ : FAILURE in 28m 12s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/6/check/gate-grenade-dsvm-ubuntu-trusty/59ceafd/ : FAILURE in 23m 17s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-nova-tox-db-functional-ubuntu-xenial/0ed0f83/ : FAILURE in 12m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/6/check/gate-grenade-dsvm-multinode/5d9cab3/ : FAILURE in 39m 50s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/cd9ce5e/ : FAILURE in 23m 24s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-cells/50cf159/ : FAILURE in 23m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph/518318f/ : FAILURE in 30m 15s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/19af9bd/ : FAILURE in 57m 37s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/1d9fa8e/ : FAILURE in 39m 11s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/04e7448/ : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":6},{"id":"a2b06d1c2bb382d34fe156766743fbabef708a99","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-09-06 19:18:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7cb7e001111cd1c49caa8ef016e406078d18a5b3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 19:27:16.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/6/ : FAILURE in 1h 28m 00s","accounts_in_message":[],"_revision_number":6},{"id":"997742e1c077bb8e74a92db755f1015290236d1a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-06 19:54:23.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/6/check-nova/Nova-ML2-Sriov/35c7bee : FAILURE in 25m 25s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/6/check-nova/Nova-MACVTAP-ML2-Sriov/3d0b5a5 : FAILURE in 27m 57s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"29a292d22879bc4de86fdf5d4f9bb09a3c8f6d05","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-06 19:57:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/6/check/dsvm-tempest-nova-network/6236c9e : SUCCESS in 1h 25m 50s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/6/check/dsvm-tempest-neutron-network/b6c9764 : SUCCESS in 1h 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a2ea9e9355659ae20a51c445e540b60bbff9cdaf","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 20:42:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/6/check/dsvm-tempest-kvm/c3ba534 : FAILURE in 55m 07s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"578fdc34f0f6232ff74510045d5bd47c4f067d8c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 21:10:14.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/6/check/check-dsvm-tempest-vz7-exe-minimal/c2b3ff0 : FAILURE in 1h 11m 08s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"c030e3452bbdb54ddbe3479e8c8f206ce261bd7b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-06 22:09:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-full-nfv/0572aad : FAILURE in 1h 08m 30s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-intel-nfv/eac7245 : FAILURE in 47m 14s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-ovsdpdk-nfv-networking/560afb2 : FAILURE in 1h 00m 02s\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-full-nfv-xenial/707911b : FAILURE in 1h 04m 25s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-intel-nfv-xenial/e517eba : FAILURE in 42m 42s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f0a3867 : FAILURE in 55m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aa33274782d27623fc9f9e79157ae4e851f640c8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-06 22:28:38.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/6/740/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"6d8a65e374547e596dc18de3b5a0a6ba0143774f","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-07 12:17:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"abecb78a700f65a43d4063ade12a5f41adf550ee","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-07 12:39:05.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/319379/7 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"1547e0d4ae7e5849b64955c24b89e4124e3d5deb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-07 12:57:18.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/7/check/check-ibm-tempest-dsvm-full/8f1e14a/ : FAILURE in 39m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"abbca55626456902d0fdf55b527ae164f377c247","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-07 13:14:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/7/check/gate-nova-docs-ubuntu-xenial/d093d4a//doc/build/html/ : SUCCESS in 2m 34s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-nova-pep8-ubuntu-xenial/4739c04/ : SUCCESS in 6m 40s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-nova-python27-db-ubuntu-xenial/7cae039/ : FAILURE in 14m 11s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/7/check/gate-nova-python34-db/9394c09/ : FAILURE in 10m 24s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/7/check/gate-nova-python35-db-nv/204d1fe/ : FAILURE in 11m 48s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/7/check/gate-grenade-dsvm-neutron-ubuntu-trusty/4a4264d/ : FAILURE in 28m 35s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/cd6494d/ : FAILURE in 46m 03s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-full-ubuntu-xenial/1db8154/ : FAILURE in 24m 51s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/7/check/gate-grenade-dsvm-ubuntu-trusty/fa601ea/ : FAILURE in 22m 55s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-nova-tox-db-functional-ubuntu-xenial/7a3613d/ : FAILURE in 14m 19s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/7/check/gate-grenade-dsvm-multinode/f791187/ : FAILURE in 43m 02s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/bd1ed00/ : FAILURE in 31m 04s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-cells/75945b3/ : FAILURE in 24m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph/9b12da2/ : FAILURE in 30m 13s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/2fbcb1c/ : FAILURE in 56m 27s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/0d55e03/ : FAILURE in 36m 12s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/7/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/359d479/ : FAILURE in 30m 10s","accounts_in_message":[],"_revision_number":7},{"id":"0a101ef62913ba23fd8ac4517d231d7cf9f1a3d9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-07 14:06:03.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/7/ : FAILURE in 1h 48m 03s","accounts_in_message":[],"_revision_number":7},{"id":"a080dc9550af2326c8dffee903d8e2165f39a482","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-07 14:41:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/7/check/dsvm-tempest-nova-network/668a972 : SUCCESS in 1h 52m 49s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/7/check/dsvm-tempest-neutron-network/1fc10cf : SUCCESS in 1h 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"457d2aecdbe33b4ea4ff7715eacc0b0d5ecb2efe","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-07 15:03:07.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/7/check-nova/Nova-ML2-Sriov/d2d3581 : FAILURE in 26m 28s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/7/check-nova/Nova-MACVTAP-ML2-Sriov/3c54241 : FAILURE in 29m 51s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"4529b82536af5043f9dddc68cf7440e17ca108fd","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-07 15:17:35.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/7/check/check-dsvm-tempest-vz7-exe-minimal/9769a5b : FAILURE in 53m 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"6eda3271f1ce7fd7057c174102f1010180102bcd","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-07 15:34:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/7/check/dsvm-tempest-kvm/63b6918 : FAILURE in 1h 07m 58s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"4cffeb155974a5a16f448a255c5ed078d13fbd03","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-07 17:04:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/7/check/tempest-dsvm-full-nfv-xenial/cdd4df9 : FAILURE in 1h 03m 59s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/7/check/tempest-dsvm-intel-nfv-xenial/28bf96a : FAILURE in 43m 14s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/b359ace : FAILURE in 56m 13s","accounts_in_message":[],"_revision_number":7},{"id":"57feb6bbace0a7e4324c4c05c38ad67fb845a23d","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-08 17:49:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"85136df03947bf94d305ab6fcfd4e94165175a2f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-08 18:25:07.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/319379/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"23bfbcf929b27d384474b04549bda10bc7808fc8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 18:34:02.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/8/ : FAILURE in 43m 05s","accounts_in_message":[],"_revision_number":8},{"id":"a9653e94adac6b31eeb834cc4643cc858d0362f1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-08 19:37:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/8/check/dsvm-tempest-xen/2c39bfa : FAILURE in 1h 46m 34s","accounts_in_message":[],"_revision_number":8},{"id":"2d6fbbeeccfdf0a35961b85f07e384e77142ee76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-08 19:46:42.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/8/check/check-ibm-tempest-dsvm-full/456d49f/ : FAILURE in 1h 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ffa3c59297b768bb2d5518d0498396731acdb7a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-08 20:49:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/8/check/gate-nova-docs-ubuntu-xenial/47bea50//doc/build/html/ : SUCCESS in 2m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-nova-pep8-ubuntu-xenial/b94ba99/ : FAILURE in 4m 29s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-nova-python27-db-ubuntu-xenial/026b79e/ : FAILURE in 36m 44s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/8/check/gate-nova-python34-db/2d9c7eb/ : FAILURE in 26m 34s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/8/check/gate-nova-python35-db-nv/e5db8b7/ : FAILURE in 36m 28s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/8/check/gate-grenade-dsvm-neutron-ubuntu-trusty/ae349c4/ : FAILURE in 51m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9c6888c/ : FAILURE in 2h 46m 43s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-full-ubuntu-xenial/7ae6f0a/ : FAILURE in 1h 57m 40s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/8/check/gate-grenade-dsvm-ubuntu-trusty/4d7d846/ : FAILURE in 35m 07s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-nova-tox-db-functional-ubuntu-xenial/39d823e/ : FAILURE in 13m 51s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/8/check/gate-grenade-dsvm-multinode/c76c85a/ : FAILURE in 49m 31s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/4147b8c/ : FAILURE in 30m 45s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-cells/3860caa/ : FAILURE in 1h 29m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph/aeb37ce/ : FAILURE in 1h 58m 12s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/c45d3ca/ : FAILURE in 2h 48m 04s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/dd8f959/ : FAILURE in 2h 20m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/8/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/1decfbe/ : FAILURE in 37m 32s","accounts_in_message":[],"_revision_number":8},{"id":"8f32b34f4d12e80a57595f77747b127cc6c6d4f9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 21:40:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/8/check/dsvm-tempest-kvm/ab758b5 : FAILURE in 1h 45m 59s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"57274da6b35c465e0a2551cf847b8b8bb5f9df64","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-08 21:54:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/8/check/dsvm-tempest-nova-network/406e234 : SUCCESS in 1h 44m 42s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/8/check/dsvm-tempest-neutron-network/938f857 : SUCCESS in 1h 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6098c183749202356b6854a2737c0aeef9358844","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 22:07:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/8/check/tempest-dsvm-full-nfv-xenial/b854c52 : FAILURE in 2h 58m 31s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/8/check/tempest-dsvm-intel-nfv-xenial/4e3a5e4 : FAILURE in 1h 03m 32s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e50035f : FAILURE in 1h 29m 03s","accounts_in_message":[],"_revision_number":8},{"id":"9ac6840c6ded9e50a5649e05975ab78e48cc84c6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-08 22:08:24.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/8/check/check-dsvm-tempest-vz7-exe-minimal/03ce806 : FAILURE in 2h 03m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"01f1fcb35e1569beb093713e7b078adf65acbd6b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-09 03:45:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/8/check-nova/Nova-ML2-Sriov/98a63db : FAILURE in 31m 46s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/8/check-nova/Nova-MACVTAP-ML2-Sriov/77b2fef : FAILURE in 28m 10s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"ea231cb1fe7759e70d5135870627535b3911937e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-09 07:45:10.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/7/807/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"b50aa917bd1faad08261fdac7183ce41232d3df5","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-09 14:54:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6c68a40c7cbd90dd19de778718bc08b8b6c13c64","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-09 15:13:24.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/319379/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"54143445753b71a8ba4f1893fc16b6ccb1d4fe03","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-09 15:34:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/9/check/dsvm-tempest-xen/f9abbef : FAILURE in 28m 01s","accounts_in_message":[],"_revision_number":9},{"id":"50751c39ece71b4290f5f1207f9b61e4a93a9623","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-09 15:41:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/9/check/tempest-dsvm-full-nfv-xenial/f0dbd51 : FAILURE in 41m 13s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/9/check/tempest-dsvm-intel-nfv-xenial/b5ee2db : FAILURE in 41m 27s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/9/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/af2884b : FAILURE in 44m 55s","accounts_in_message":[],"_revision_number":9},{"id":"b2ae9883eefae777f86aadd603703902175936f2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-09 15:43:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/9/check-nova/Nova-ML2-Sriov/dc9d5df : FAILURE in 3m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/9/check-nova/Nova-MACVTAP-ML2-Sriov/334d097 : FAILURE in 2m 46s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"014aaf8d438afa4431e69647f32c55c7976867b6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-09 15:53:05.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/9/ : FAILURE in 57m 47s","accounts_in_message":[],"_revision_number":9},{"id":"1fede81b65ec493cefd0c815e4bf1406c01b5732","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-09 16:03:17.000000000","message":"Patch Set 9:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/9/check/check-ibm-tempest-dsvm-full/ef7036b/ : SUCCESS in 1h 07m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"67a2e13eccc207f2f3226ba04fb5f4d9d573ebf5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-09 16:20:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/9/check/dsvm-tempest-nova-network/cd0c970 : SUCCESS in 1h 24m 19s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/9/check/dsvm-tempest-neutron-network/4395c6e : FAILURE in 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3035967ca3dcee6f49b0178dd1d0b265fd707695","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-09 16:28:45.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/9/check/dsvm-tempest-kvm/ab3f317 : SUCCESS in 1h 07m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"b3dfee678f652263229f109a64787d1491b234f4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-09 16:50:36.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/9/check/check-dsvm-tempest-vz7-exe-minimal/0cf67db : FAILURE in 1h 16m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"0158ec1e4878bacfe329777ff23ef7a169342119","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-09 16:53:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/9/check/gate-nova-docs-ubuntu-xenial/89b6285//doc/build/html/ : SUCCESS in 5m 20s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-nova-pep8-ubuntu-xenial/a616711/ : SUCCESS in 6m 29s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-nova-python27-db-ubuntu-xenial/7b7f7df/ : FAILURE in 25m 01s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/9/check/gate-nova-python34-db/36ea0c7/ : FAILURE in 25m 20s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/9/check/gate-nova-python35-db-nv/c4e94fa/ : FAILURE in 22m 18s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/9/check/gate-grenade-dsvm-neutron-ubuntu-trusty/dbbe7c0/ : FAILURE in 29m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5bd6ff8/ : FAILURE in 1h 08m 41s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-full-ubuntu-xenial/c408d71/ : FAILURE in 44m 43s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/9/check/gate-grenade-dsvm-ubuntu-trusty/d676342/ : FAILURE in 32m 45s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-nova-tox-db-functional-ubuntu-xenial/06fe309/ : FAILURE in 14m 07s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/9/check/gate-grenade-dsvm-multinode/0e18817/ : FAILURE in 33m 23s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/442cb30/ : SUCCESS in 24m 07s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-cells/3ba5c61/ : FAILURE in 1h 56m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-full-devstack-plugin-ceph/b7ac8e9/ : FAILURE in 52m 24s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-trusty http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-trusty/2e3894e/ : FAILURE in 1h 24m 31s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/b15b528/ : FAILURE in 1h 37m 46s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/254874b/ : FAILURE in 1h 29m 32s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/9/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/fc155d4/ : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":9},{"id":"4765428420f4686b1f9e402d9692490b3f6b48fc","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-09 17:20:28.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-multinode: https://bugs.launchpad.net/bugs/1621916\n- gate-grenade-dsvm-neutron-ubuntu-trusty: unrecognized error\n- gate-grenade-dsvm-ubuntu-trusty: unrecognized error\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python34-db: unrecognized error\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph: unrecognized error\n- gate-tempest-dsvm-full-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-trusty: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"6b3e69e6d8298dcacd7fbfae4a8b9b935ce5ad1e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-10 08:48:06.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/9/855/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"940e3e7be1415c3db548260d92d67595f5777a6c","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-12 09:59:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"51a8d5e3aee016a567de918019688d47ee87c4d8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-12 10:25:53.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/319379/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"133d8a4aac81e1da360d352f02315fe1a9ce4290","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-12 10:36:16.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/10/ : FAILURE in 35m 54s","accounts_in_message":[],"_revision_number":10},{"id":"6cdfafbe780ea743a841df6e64f7ca3cdf950e72","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-12 11:04:11.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/10/check/check-ibm-tempest-dsvm-full/bbe413e/ : FAILURE in 1h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bd747111f781f2966b566b262a8ce1700bcb7ee8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-12 11:25:36.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/10/907/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"bcc571b3d78e032e28c77b3a9d1998a2bf196647","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-12 11:38:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/10/check/dsvm-tempest-xen/927ccc3 : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":10},{"id":"d9cf4aef16c6bee65efb34ff31f1fcba7a60de9b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-12 11:58:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/10/check/gate-nova-docs-ubuntu-xenial/0e38a7e//doc/build/html/ : SUCCESS in 2m 46s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-nova-pep8-ubuntu-xenial/f5bdeca/ : SUCCESS in 4m 42s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-nova-python27-db-ubuntu-xenial/fe888e4/ : FAILURE in 34m 56s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/10/check/gate-nova-python34-db/c841d20/ : FAILURE in 25m 17s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/10/check/gate-nova-python35-db-nv/bda3001/ : FAILURE in 24m 16s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/10/check/gate-grenade-dsvm-neutron-ubuntu-trusty/e072722/ : FAILURE in 36m 06s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8dbcf63/ : FAILURE in 1h 25m 04s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-full-ubuntu-xenial/ec3d4ef/ : SUCCESS in 55m 37s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/10/check/gate-grenade-dsvm-ubuntu-trusty/e9ed346/ : FAILURE in 31m 33s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-nova-tox-db-functional-ubuntu-xenial/3c5eb33/ : FAILURE in 15m 41s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/10/check/gate-grenade-dsvm-multinode/2036abf/ : FAILURE in 39m 44s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/b27ea17/ : SUCCESS in 24m 24s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-cells/0f6f30d/ : FAILURE in 1h 56m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-full-devstack-plugin-ceph/b016350/ : FAILURE in 51m 01s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/3ae4672/ : FAILURE in 1h 15m 32s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/345f688/ : FAILURE in 1h 28m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/10/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c036ceb/ : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":10},{"id":"99b6cdc5f77858c9b181c3ab8d63cbfa9bb35304","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-12 12:00:48.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-multinode: unrecognized error\n- gate-grenade-dsvm-neutron-ubuntu-trusty: unrecognized error\n- gate-grenade-dsvm-ubuntu-trusty: unrecognized error\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python34-db: unrecognized error\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"ccda8d3e95fba626c0ed590897a1cdb7e5281b14","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-12 12:12:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/10/check-nova/Nova-ML2-Sriov/444d103 : SUCCESS in 34m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/10/check-nova/Nova-MACVTAP-ML2-Sriov/d2c0749 : SUCCESS in 33m 30s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"0662c46971a83536013bb89972c2568c2a8d110a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-12 12:23:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/10/check/tempest-dsvm-full-nfv-xenial/a2f8993 : FAILURE in 1h 59m 57s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/10/check/tempest-dsvm-intel-nfv-xenial/0f2e373 : FAILURE in 54m 20s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a721307 : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":10},{"id":"2e77925aca5e76d41b2e7ee1e5da1b005e73b0c3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-12 12:45:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/10/check/dsvm-tempest-nova-network/a2d0c67 : SUCCESS in 1h 20m 51s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/10/check/dsvm-tempest-neutron-network/705015d : SUCCESS in 1h 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8236bf547562d2cac2b24a904c7b2beb0f9183e3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-12 14:05:14.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/10/check/dsvm-tempest-kvm/14b16e2 : FAILURE in 1h 20m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"4c3dd29053515099374d0b7d1f82473f2fae1a87","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-12 14:20:34.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/10/check/check-dsvm-tempest-vz7-exe-minimal/51fc4a8 : FAILURE in 2h 01m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"9d7e6127318125313c7f42bf262843d4a3ad4bcb","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-15 16:31:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2cfc093f0f8f7efe775728b734f45ffb49fd907e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 17:08:45.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/11/ : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":11},{"id":"876c704504b08d87438af36a100ad90be4917e00","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-15 18:28:35.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/11/check/check-ibm-tempest-dsvm-full/f818c12/ : FAILURE in 1h 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"71a0f8d208d654a22d8614983f766b57f2c0c2db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-15 19:20:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/11/check/gate-nova-docs-ubuntu-xenial/70b1dca//doc/build/html/ : SUCCESS in 3m 35s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-nova-pep8-ubuntu-xenial/bec0985/ : SUCCESS in 6m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-nova-python27-db-ubuntu-xenial/e57f7a4/ : FAILURE in 24m 32s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/11/check/gate-nova-python34-db/8a8cd92/ : FAILURE in 28m 08s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/11/check/gate-nova-python35-db-nv/813cea3/ : FAILURE in 27m 56s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/11/check/gate-grenade-dsvm-neutron-ubuntu-trusty/683a6ff/ : FAILURE in 30m 27s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f0000b9/ : FAILURE in 2h 46m 53s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-full-ubuntu-xenial/b8edebf/ : FAILURE in 2h 45m 32s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/11/check/gate-grenade-dsvm-ubuntu-trusty/412aeef/ : FAILURE in 24m 29s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-nova-tox-db-functional-ubuntu-xenial/85f9d5b/ : FAILURE in 13m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/11/check/gate-grenade-dsvm-multinode/6b09b3f/ : FAILURE in 42m 24s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-cells-ubuntu-xenial/0ace7a3/ : FAILURE in 1h 56m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ee7ae41/ : FAILURE in 2h 46m 41s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/b3ae710/ : FAILURE in 2h 48m 05s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/adc7df6/ : FAILURE in 2h 47m 05s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/89d7c0d/ : FAILURE in 34m 55s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/11/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/aebe6ed/ : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"17c93a519406b325b410b8ec0832a05a6b77546a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-15 20:29:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/11/check/dsvm-tempest-nova-network/38c0923 : SUCCESS in 1h 25m 52s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/11/check/dsvm-tempest-neutron-network/b0df572 : FAILURE in 1h 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2dbea9fe645e10bf09643920ede4f88a95a8158b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 20:51:10.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/11/check/check-dsvm-tempest-vz7-exe-minimal/3a59a0c : FAILURE in 2h 16m 55s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"98055efb30e3ecb66629e9e592dd4e6f4a9097e7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 22:10:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/11/check/tempest-dsvm-full-nfv-xenial/a4d6ad5 : FAILURE in 2h 59m 18s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/11/check/tempest-dsvm-intel-nfv-xenial/f715391 : FAILURE in 1h 12m 09s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/77f2506 : FAILURE in 1h 29m 05s","accounts_in_message":[],"_revision_number":11},{"id":"32bbbaa53cfbd570e1177526d1994678d692fcc2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-15 22:12:03.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/11/check-nova/Nova-ML2-Sriov/428ee8a : SUCCESS in 34m 10s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/11/check-nova/Nova-MACVTAP-ML2-Sriov/b609a7a : SUCCESS in 33m 53s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"fc15da152ea5eb9110a2d3b00a8acb5d5151861c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-16 03:16:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/11/1103/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"a98d49b5bde3f19e820bc3fb3ff61b7f5dd92843","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-19 17:33:03.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"46e863c8b1c92904abbf2205ee94c03b19986c92","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-19 17:48:02.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/319379/12 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"44953c51b3346e23c8e27ab19bd742f9ba73c288","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-19 18:21:53.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/12/check/check-ibm-tempest-dsvm-full/6b076f4/ : FAILURE in 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1d3a9303ea964d3eb39d1eda93621599529aa3ae","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-19 18:38:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/12/check/dsvm-tempest-xen/621d592 : FAILURE in 53m 58s","accounts_in_message":[],"_revision_number":12},{"id":"9b5a21a0217c5cbd06e5d5dd1166169bb7ab6616","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 18:53:06.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/12/ : FAILURE in 1h 18m 58s","accounts_in_message":[],"_revision_number":12},{"id":"06a9415e0d529284f85faaf03a5701c36b16d85c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-19 18:55:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/12/check/gate-nova-docs-ubuntu-xenial/a899930//doc/build/html/ : SUCCESS in 2m 32s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-nova-pep8-ubuntu-xenial/8a91b5f/ : FAILURE in 5m 15s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-nova-python27-db-ubuntu-xenial/7148c09/ : FAILURE in 37m 14s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/12/check/gate-nova-python34-db/c3a87c4/ : FAILURE in 23m 01s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/12/check/gate-nova-python35-db-nv/04d4562/ : FAILURE in 26m 10s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/12/check/gate-grenade-dsvm-neutron-ubuntu-trusty/abc2c6e/ : FAILURE in 45m 40s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7ce097f/ : FAILURE in 39m 35s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-full-ubuntu-xenial/852588e/ : FAILURE in 32m 21s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/12/check/gate-grenade-dsvm-ubuntu-trusty/6ef8b70/ : FAILURE in 32m 52s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-nova-tox-db-functional-ubuntu-xenial/2e024fd/ : FAILURE in 12m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/12/check/gate-grenade-dsvm-multinode/364ea11/ : FAILURE in 42m 52s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-cells-ubuntu-xenial/fde4163/ : FAILURE in 1h 18m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/3e2b1a5/ : FAILURE in 33m 29s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/78b9ca9/ : FAILURE in 51m 00s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/0e1f15b/ : FAILURE in 43m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d8d7f41/ : FAILURE in 17m 08s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/12/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/48d184c/ : FAILURE in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4b595b544f38e52937b53f4328b6605595a664e4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-19 19:01:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/12/check-nova/Nova-ML2-Sriov/e3055fe : SUCCESS in 33m 44s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/12/check-nova/Nova-MACVTAP-ML2-Sriov/4281bfd : SUCCESS in 33m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"c8e860cb101a0d9acf3d9b1c6fdfb6522ccd6155","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-19 20:02:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/12/check/dsvm-tempest-nova-network/78191cf : FAILURE in 40m 51s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/12/check/dsvm-tempest-neutron-network/b5414ce : SUCCESS in 1h 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fd92b9239b882030bd397c8be8eba28c8a5a6b4e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 20:31:35.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/12/check/tempest-dsvm-full-nfv-xenial/4c2b5fd : FAILURE in 1h 20m 16s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/12/check/tempest-dsvm-intel-nfv-xenial/cbe00f3 : FAILURE in 34m 45s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0211697 : FAILURE in 48m 19s","accounts_in_message":[],"_revision_number":12},{"id":"6686f16dfad7fdb1a803ca14908e74843a1f48ea","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-20 13:25:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/12/check/dsvm-tempest-nova-network/234e9b7 : SUCCESS in 1h 47m 39s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/12/check/dsvm-tempest-neutron-network/01f756d : SUCCESS in 1h 48m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"40ee9b4f458fad173aa477d0e8bc51a7fbb4bb64","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-21 16:03:45.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"96cb2e9714107edb8c5e1a559dd0131e465efdb0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-21 16:53:39.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/319379/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"40fc0f17c61bb2c97eddf58eea418c1a0a0c7eb6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-21 16:54:44.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/13/check/check-ibm-tempest-dsvm-full/b3f47fb/ : FAILURE in 49m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"87b1b1fb147d84b1fdc18f7e1cd2ca1179a454de","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-21 17:15:48.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/13/check/dsvm-tempest-xen/3471a09 : FAILURE in 1h 00m 24s","accounts_in_message":[],"_revision_number":13},{"id":"a9ac45971c2e5448d708d614f8a1ef24f1202682","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-21 18:26:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/13/check/gate-nova-docs-ubuntu-xenial/c4a93ed//doc/build/html/ : SUCCESS in 3m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-nova-pep8-ubuntu-xenial/26c2239/ : FAILURE in 4m 54s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-nova-python27-db-ubuntu-xenial/3f77ffa/ : FAILURE in 36m 13s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/13/check/gate-nova-python34-db/de367af/ : FAILURE in 25m 27s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/13/check/gate-nova-python35-db-nv/5d84ef9/ : FAILURE in 27m 59s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/13/check/gate-grenade-dsvm-neutron-ubuntu-trusty/6850ad7/ : FAILURE in 49m 42s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/370aabb/ : FAILURE in 46m 39s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-full-ubuntu-xenial/5bac425/ : FAILURE in 26m 55s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/13/check/gate-grenade-dsvm-ubuntu-trusty/c0e5e01/ : FAILURE in 31m 00s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-nova-tox-db-functional-ubuntu-xenial/75471e2/ : FAILURE in 15m 22s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/13/check/gate-grenade-dsvm-multinode/70f302e/ : FAILURE in 44m 00s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-cells-ubuntu-xenial/8402a2e/ : FAILURE in 2h 15m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/49c40d5/ : FAILURE in 34m 48s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/dc86431/ : FAILURE in 58m 34s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/2742f1b/ : FAILURE in 38m 33s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/2bffb8c/ : FAILURE in 17m 00s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/13/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/c0998dc/ : FAILURE in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6e5af5735bf8b1c9fb0e5fd0af7dc10047d4330d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-21 18:47:14.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/13/ : FAILURE in 2h 07m 14s","accounts_in_message":[],"_revision_number":13},{"id":"4ce0a6252bfe09f31b592bd03cbcb70bb7989809","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-21 20:18:37.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/13/check-nova/Nova-ML2-Sriov/c67f377 : FAILURE in 25m 48s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/13/check-nova/Nova-MACVTAP-ML2-Sriov/b1141f5 : FAILURE in 25m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"3f95c51e1ac6803ea7b641ea7cf05d15856a84fe","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-21 22:07:37.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/13/check/dsvm-tempest-nova-network/53b1e1f : SUCCESS in 1h 43m 03s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/13/check/dsvm-tempest-neutron-network/b885bd8 : SUCCESS in 1h 54m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"dc5f6a5ea25d6f3190789348b2627f02e4b14ad7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-22 00:21:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/13/check/tempest-dsvm-full-nfv-xenial/9e4de66 : FAILURE in 1h 21m 39s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/13/check/tempest-dsvm-intel-nfv-xenial/0520a17 : FAILURE in 43m 25s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/13/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/d7a0184 : FAILURE in 1h 02m 45s","accounts_in_message":[],"_revision_number":13},{"id":"ccfe0a47288efa555a8424e55535c957f7f45462","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-22 14:32:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"caa39e9c1950565bb4fd9b0ad98386f28af474e1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-22 14:32:58.000000000","message":"Patch Set 14:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"92640aca213d9b9cf48b74178abcc7addcf4b045","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-22 15:03:19.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/319379/14 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"e37d347dde5cf1d3a8c192f8dd0214776f6812ad","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-22 15:51:48.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/14/check/dsvm-tempest-xen/1afa1c5 : FAILURE in 58m 25s","accounts_in_message":[],"_revision_number":14},{"id":"2be00bc25b80cb1e0dac0de1d0ae061a7372cb2c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-22 16:30:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/14/check/gate-nova-docs-ubuntu-xenial/f22630f//doc/build/html/ : SUCCESS in 3m 26s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-nova-pep8-ubuntu-xenial/3fe09c5/ : FAILURE in 5m 42s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-nova-python27-db-ubuntu-xenial/a584ffd/ : FAILURE in 33m 14s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/14/check/gate-nova-python34-db/aa0bb95/ : FAILURE in 29m 53s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/14/check/gate-nova-python35-db-nv/8517cc4/ : FAILURE in 33m 16s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/79/319379/14/check/gate-grenade-dsvm-neutron-ubuntu-trusty/00bed1b/ : FAILURE in 1h 10m 25s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c5237f3/ : FAILURE in 38m 51s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-full-ubuntu-xenial/0ffdb46/ : FAILURE in 27m 49s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/79/319379/14/check/gate-grenade-dsvm-ubuntu-trusty/37107f4/ : FAILURE in 42m 22s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-nova-tox-db-functional-ubuntu-xenial/25ae2b4/ : FAILURE in 13m 14s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/14/check/gate-grenade-dsvm-multinode/78e1412/ : FAILURE in 26m 53s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-cells-ubuntu-xenial/f0fecf0/ : FAILURE in 1h 42m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/75fda8f/ : FAILURE in 39m 27s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/1bc3c57/ : FAILURE in 57m 43s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/c02e4de/ : FAILURE in 1h 20m 10s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/1c52e07/ : FAILURE in 18m 11s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/14/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/ff5a0e4/ : FAILURE in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"97274439b0ccb6cda8901c50cdf13007c5f2f0ff","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-22 17:20:57.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/14/ : FAILURE in 2h 27m 51s","accounts_in_message":[],"_revision_number":14},{"id":"165e9bc0c2e0ef6e7fb49c7b8b33c03803a15fcb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-22 19:36:23.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/14/check-nova/Nova-ML2-Sriov/9e43dd5 : FAILURE in 27m 05s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/14/check-nova/Nova-MACVTAP-ML2-Sriov/c66c9f1 : FAILURE in 27m 31s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":14},{"id":"e8c6c4d2e81b2e7da2e47fbc23b826118db72ede","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-22 21:07:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/14/check/tempest-dsvm-full-nfv-xenial/b822c10 : FAILURE in 1h 27m 20s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/14/check/tempest-dsvm-intel-nfv-xenial/388cd5f : FAILURE in 46m 47s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/57f06c3 : FAILURE in 52m 06s","accounts_in_message":[],"_revision_number":14},{"id":"5f992acd22385ed2c37403ff4b044fb1b955a027","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-23 06:44:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/14/check/dsvm-tempest-nova-network/27233f0 : SUCCESS in 2h 02m 09s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/14/check/dsvm-tempest-neutron-network/a6906a7 : SUCCESS in 1h 45m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c026d5e24263e006ad86b783b29446f12ab5e624","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-24 06:53:39.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/14/1261/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"329da9b7ebacc29e42b7aad93bf4357ed8a94645","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-26 08:58:32.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"5e5ceb72c14c7ec378c31b0418b0a58bca1a1842","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-26 08:59:22.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a8159d754539f7748dc767a9af5ec26b6578ab63","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-26 09:25:04.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/319379/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"0252086b655aee43b66f246239418cfb247f42a9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-26 10:04:13.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/15/check/dsvm-tempest-xen/ba70888 : FAILURE in 52m 30s","accounts_in_message":[],"_revision_number":15},{"id":"26d9b63c54201bb8cdbcc3269a5a38ae694d1a9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 10:22:10.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/15/check/gate-nova-docs-ubuntu-xenial/057aaa7//doc/build/html/ : SUCCESS in 4m 28s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-nova-pep8-ubuntu-xenial/32e6e8c/ : FAILURE in 4m 57s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-nova-python27-db-ubuntu-xenial/c9fa293/ : FAILURE in 25m 34s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/15/check/gate-nova-python34-db/d95d45b/ : FAILURE in 18m 08s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/15/check/gate-nova-python35-db-nv/03f4f76/ : FAILURE in 22m 17s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/93d7e3c/ : FAILURE in 35m 54s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-grenade-dsvm-neutron-ubuntu-xenial/32c4f4c/ : FAILURE in 39m 51s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-full-ubuntu-xenial/58e843d/ : FAILURE in 21m 33s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-grenade-dsvm-ubuntu-xenial/aa34bb0/ : FAILURE in 23m 55s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-nova-tox-db-functional-ubuntu-xenial/622382f/ : FAILURE in 12m 57s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/15/check/gate-grenade-dsvm-multinode/d45c1e2/ : FAILURE in 43m 53s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-cells-ubuntu-xenial/b62ed02/ : FAILURE in 1h 20m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e4703e8/ : FAILURE in 40m 59s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/ec737b2/ : FAILURE in 55m 08s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/5146fa8/ : FAILURE in 31m 17s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/4ab5dd5/ : FAILURE in 21m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/15/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/82b4ea6/ : FAILURE in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b7707388f6864e273ca74f621b7c9af98f88a749","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-26 10:55:04.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/15/check/tempest-dsvm-full-nfv-xenial/a49b7a7 : FAILURE in 1h 32m 34s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/15/check/tempest-dsvm-intel-nfv-xenial/e56ab66 : FAILURE in 43m 07s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/816bf5b : FAILURE in 1h 01m 56s","accounts_in_message":[],"_revision_number":15},{"id":"742688053957658236b66d9ddf738085ce22df1c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-26 10:58:40.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/15/check/dsvm-tempest-nova-network/b235ac0 : FAILURE in 45m 37s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/15/check/dsvm-tempest-neutron-network/b93a9cd : FAILURE in 1h 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ea002f119bc8eb4ebb8841d98992b22fec3932dc","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-26 11:05:49.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/15/ : FAILURE in 2h 05m 56s","accounts_in_message":[],"_revision_number":15},{"id":"9bb0f174df2e82cde1b4c5007243adbe04ebeab0","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-26 16:14:51.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"fa2bddf4f08573da89db8b526f4f8a96d6065fa8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-26 16:36:20.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/pci/319379/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"b81c0851530e3da81211be88b285abd7fe9574ea","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-26 17:03:15.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/16/check/dsvm-tempest-kvm/18d6243 : SUCCESS in 46m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"ba70d78289d39fad9a8d0ad76e69b405e0718dda","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-26 17:21:00.000000000","message":"Patch Set 16:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/16/check/check-ibm-tempest-dsvm-full/f1fc243/ : FAILURE in 1h 05m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"79ab9fa279d85a383e5e6d229f46ef615d06d7d0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-26 17:56:16.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/16/ : FAILURE in 1h 40m 25s","accounts_in_message":[],"_revision_number":16},{"id":"5025597782b2f70c897eefe18ca8b2b18a57315f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-26 17:57:18.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/16/check/dsvm-tempest-nova-network/c6c55ce : FAILURE in 1h 17m 05s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/16/check/dsvm-tempest-neutron-network/7d9cde5 : SUCCESS in 1h 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"eb7600318fac37ed00eb850fdcafcf9ac205f5a5","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-26 17:58:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/16/check/dsvm-tempest-xen/88d4b85 : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":16},{"id":"5fbbbedac50105ee393abafd44220165d4a633fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-26 18:24:05.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/16/check/gate-nova-docs-ubuntu-xenial/c73aef3//doc/build/html/ : SUCCESS in 3m 37s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-nova-pep8-ubuntu-xenial/1f7f1e5/ : FAILURE in 7m 03s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-nova-python27-db-ubuntu-xenial/0675f1e/ : FAILURE in 17m 44s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/16/check/gate-nova-python34-db/53c9b58/ : FAILURE in 9m 01s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/16/check/gate-nova-python35-db-nv/439284d/ : FAILURE in 12m 55s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/b455cf7/ : FAILURE in 59m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c5d523c/ : FAILURE in 53m 06s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-full-ubuntu-xenial/8b8d8ec/ : FAILURE in 1h 09m 51s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-grenade-dsvm-ubuntu-xenial/6f04fed/ : SUCCESS in 1h 14m 29s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-nova-tox-db-functional-ubuntu-xenial/d3597b0/ : SUCCESS in 13m 29s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/16/check/gate-grenade-dsvm-multinode/ca9d8c5/ : SUCCESS in 43m 02s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-cells-ubuntu-xenial/6a3477d/ : FAILURE in 36m 39s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a2b137a/ : SUCCESS in 1h 12m 35s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/3cf3d43/ : FAILURE in 18m 57s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/a30c869/ : FAILURE in 1h 18m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/e5aaa4b/ : SUCCESS in 23m 48s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/16/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/c1e0a30/ : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"011c99b17c60fc6cb79369001d72c8bf7f940ac1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-26 20:11:53.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-full-nfv-xenial/8926726 : FAILURE in 1h 50m 21s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-intel-nfv-xenial/59739d6 : FAILURE in 33m 45s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/79df653 : FAILURE in 36m 41s","accounts_in_message":[],"_revision_number":16},{"id":"beb2476044709a636405ea949b761a3755ae7e26","author":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"date":"2016-09-27 11:35:45.000000000","message":"Patch Set 16:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":16},{"id":"2c4750e447c2400e6e913b558d565babdc3dacb1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-27 19:23:14.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-full-nfv-xenial/0b15077 : FAILURE in 1h 46m 13s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-intel-nfv-xenial/608bf27 : FAILURE in 40m 45s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/16/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/906120d : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":16},{"id":"abaf507cf1bc15fb297c9dcf83f1609fe16cdf8d","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-28 10:57:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5710e6591faff890c1699f9bae678ad15a02b9a2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-28 11:18:26.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/319379/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"bd06f30a1db5073ceadf6f52f1c0d7a72e4ee576","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-28 12:04:14.000000000","message":"Patch Set 17:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/17/check/check-ibm-tempest-dsvm-full/51d9316/ : FAILURE in 1h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2be55e0bc063c80a4e04e6a73e93c14037281088","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-28 12:29:53.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/17/check/gate-nova-docs-ubuntu-xenial/a6ce99a//doc/build/html/ : SUCCESS in 3m 10s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-nova-pep8-ubuntu-xenial/ee55f00/ : SUCCESS in 5m 21s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-nova-python27-db-ubuntu-xenial/caf516d/ : FAILURE in 9m 42s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/17/check/gate-nova-python34-db/69d0b5c/ : FAILURE in 8m 04s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/17/check/gate-nova-python35-db-nv/f64218b/ : FAILURE in 9m 16s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bff638a/ : FAILURE in 50m 46s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-grenade-dsvm-neutron-ubuntu-xenial/5b498f5/ : FAILURE in 30m 30s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-full-ubuntu-xenial/e093054/ : FAILURE in 36m 02s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-grenade-dsvm-ubuntu-xenial/99109f1/ : SUCCESS in 24m 54s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-nova-tox-db-functional-ubuntu-xenial/0ce6dd1/ : SUCCESS in 9m 40s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/17/check/gate-grenade-dsvm-multinode/d673625/ : SUCCESS in 42m 30s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-cells-ubuntu-xenial/b3d8926/ : FAILURE in 1h 13m 47s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/77d7159/ : FAILURE in 40m 44s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/87a0e27/ : FAILURE in 1h 20m 01s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/8094ed0/ : FAILURE in 49m 51s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0abfa36/ : SUCCESS in 17m 02s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/17/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/70aeec3/ : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9c4d9ee9deb104df530bb98769df395b490a304f","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-28 12:30:13.000000000","message":"Patch Set 17:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-neutron-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python34-db: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":17},{"id":"15f86c1622495f1d95ea336c2729ad7bb26460bf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-28 12:48:29.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/17/check/dsvm-tempest-xen/ea88373 : FAILURE in 1h 36m 42s","accounts_in_message":[],"_revision_number":17},{"id":"ad4bb6c9896fd0065a7478c0d3e8bab50ddcdfe7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-28 13:00:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/17/check/dsvm-tempest-nova-network/4423d84 : SUCCESS in 1h 24m 22s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/17/check/dsvm-tempest-neutron-network/b38d08f : SUCCESS in 1h 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"cbfae28ef383f3a464ee86b971f2a296be35b757","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-28 15:02:00.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/17/ : FAILURE in 4h 00m 02s","accounts_in_message":[],"_revision_number":17},{"id":"558fd119a78eb720a941a257df42da988649db80","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-28 16:10:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/17/check/tempest-dsvm-full-nfv-xenial/f66ced8 : FAILURE in 1h 56m 35s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/17/check/tempest-dsvm-intel-nfv-xenial/013c811 : FAILURE in 43m 08s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/17/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/43a88e9 : SUCCESS in 1h 03m 05s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6ebdcd8edd27922bb54efc8465d93611baf1b03b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-29 02:36:56.000000000","message":"Patch Set 16: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/16/1437/ : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"e164e68c738f92728f72dd15d5b36d5f4e2a8021","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-29 02:41:24.000000000","message":"Patch Set 16: -Verified\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/16/1439/ : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"bf2709a33990bd0c934cfcb51b84977724a5c3fb","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-29 09:40:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"649619ffa0d8ccff855da46edc8c089dd491f471","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-09-29 09:49:28.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"b53574167fda742ab9eeb485b3947b10f39d4210","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-29 10:22:42.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/319379/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"1c49e436cc727d59dd1d9a5474ae425394c54d7b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-29 11:08:37.000000000","message":"Patch Set 19:\n\n* pci-test http://52.27.155.124/pci/319379/19 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"cb911d92bf314c040c2a044f266f45327cd9d778","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-29 11:44:11.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/19/check/dsvm-tempest-xen/1ea8023 : FAILURE in 1h 47m 56s","accounts_in_message":[],"_revision_number":19},{"id":"43f8b9f3382c3dc990e2893facf007140fb1e96e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-29 12:11:21.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/19/check/gate-nova-docs-ubuntu-xenial/a511a09//doc/build/html/ : SUCCESS in 2m 32s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-nova-pep8-ubuntu-xenial/70e9427/ : SUCCESS in 4m 30s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-nova-python27-db-ubuntu-xenial/9b6f8cc/ : FAILURE in 9m 37s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/19/check/gate-nova-python34-db/a0fd444/ : FAILURE in 8m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/19/check/gate-nova-python35-db-nv/14aaa45/ : FAILURE in 8m 58s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9253ea5/ : FAILURE in 2h 12m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e2a921a/ : FAILURE in 43m 43s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-full-ubuntu-xenial/732f17a/ : FAILURE in 1h 58m 38s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-grenade-dsvm-ubuntu-xenial/ab117b3/ : FAILURE in 39m 38s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-nova-tox-db-functional-ubuntu-xenial/54b9f88/ : FAILURE in 9m 49s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/19/check/gate-grenade-dsvm-multinode/bf07a71/ : FAILURE in 38m 53s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-cells-ubuntu-xenial/f9fa647/ : FAILURE in 1h 35m 30s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7021194/ : FAILURE in 1h 56m 49s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f14d57c/ : FAILURE in 18m 38s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f59a30a/ : FAILURE in 2h 13m 25s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b9b99e8/ : FAILURE in 33m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/19/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/3166ba4/ : FAILURE in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c7a6cd0364878ea3d5f2f4fbf1205c3697543923","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-29 13:57:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/19/check/dsvm-tempest-nova-network/2f030dd : FAILURE in 2h 27m 40s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/19/check/dsvm-tempest-neutron-network/77df6e2 : FAILURE in 3h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"37b3e1c9e4e5a24ef485e86c251a3fe3e01285ad","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-29 14:56:24.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/19/check/tempest-dsvm-full-nfv-xenial/4ae084b : FAILURE in 2h 48m 56s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/19/check/tempest-dsvm-intel-nfv-xenial/fa4a9d1 : FAILURE in 1h 07m 33s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/19/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9f9d187 : FAILURE in 1h 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c72c0d2e42e9fe4612fd75877a8d3d4fcdb11b6b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-29 23:12:54.000000000","message":"Patch Set 19:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/19/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":19},{"id":"809b57de85bb0e722ebd9ce3c2323f2b869be452","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-30 15:15:15.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/19/1562/ : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"0a96ee37082c95b1f0aa2b557abbb99c376331d4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-01 03:32:13.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/319379/17/1627/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"5b5bc9ee629cf334c7c7df290156bd2d04113a34","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-10-04 14:14:45.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"12bd87dfa3709a43b33c0addd43c50e5e22099c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-04 14:17:25.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"df0ebd4245e1c2d9bd0b593e50f0e48fc13ad542","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-10-04 14:36:27.000000000","message":"Patch Set 20:\n\n* pci-test http://52.27.155.124/pci/319379/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"b7e5d09e8d6ce0c1ccd9f07a1dc9f2e8ec3c8856","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-04 14:51:34.000000000","message":"Patch Set 20: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/20/38/ : SUCCESS\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":20},{"id":"44482d48915f6a5f94327bb5ee93bf140364b30f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-04 14:54:20.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/20/ : FAILURE in 37m 11s","accounts_in_message":[],"_revision_number":20},{"id":"aa5b32deedacbf1ae4369e732505c282df55e46e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-04 15:46:35.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/20/check/gate-nova-docs-ubuntu-xenial/6456ab1//doc/build/html/ : SUCCESS in 3m 12s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-nova-pep8-ubuntu-xenial/a72e152/ : SUCCESS in 7m 44s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-nova-python27-db-ubuntu-xenial/6b59bd4/ : FAILURE in 9m 37s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/20/check/gate-nova-python34-db/883e737/ : FAILURE in 10m 51s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/20/check/gate-nova-python35-db-nv/cdfdac7/ : FAILURE in 9m 27s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6219ee0/ : FAILURE in 55m 09s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-grenade-dsvm-neutron-ubuntu-xenial/44fe85e/ : FAILURE in 31m 39s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-full-ubuntu-xenial/d44d08e/ : FAILURE in 55m 38s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-grenade-dsvm-ubuntu-xenial/0b34dce/ : SUCCESS in 26m 29s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-nova-tox-db-functional-ubuntu-xenial/f06946e/ : SUCCESS in 10m 04s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/79/319379/20/check/gate-grenade-dsvm-multinode/efaa7e9/ : SUCCESS in 27m 38s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-cells-ubuntu-xenial/325126d/ : FAILURE in 1h 27m 58s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e08904c/ : FAILURE in 40m 20s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/a51d4b9/ : FAILURE in 18m 16s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f2cd8d7/ : FAILURE in 52m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/9a6219c/ : SUCCESS in 19m 29s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/20/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/25f4207/ : SUCCESS in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3d2ba376f5c585282646f2ede4e9435ff38e34d8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-10-04 16:23:34.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/20/check-nova/Nova-ML2-Sriov/365ab0f : SUCCESS in 33m 09s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/20/check-nova/Nova-MACVTAP-ML2-Sriov/cf60b9d : SUCCESS in 35m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":20},{"id":"e76e255ff06d433b5be3cd06e5beef45cae3b9fc","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-10-04 17:02:46.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/20/check/dsvm-tempest-xen/42ce5cd : FAILURE in 1h 37m 53s","accounts_in_message":[],"_revision_number":20},{"id":"ed26c82c55751b2637e992d473442f28d4e36536","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-10-04 17:08:15.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/20/check/tempest-dsvm-full-nfv-xenial/7508a1b : FAILURE in 1h 51m 29s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/20/check/tempest-dsvm-intel-nfv-xenial/adea277 : FAILURE in 43m 09s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/20/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/72b635e : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":20},{"id":"3a5225185f6bdd9bff370539f81c066e409360fb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-10-07 09:48:09.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/20/check/dsvm-tempest-nova-network/2ca0584 : SUCCESS in 1h 19m 02s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/20/check/dsvm-tempest-neutron-network/1157401 : SUCCESS in 1h 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"15b516e88099825f87fce658277a9863597f20bb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-02 14:28:50.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"880502110ac93a0fb222a90f5584bdfe5323e9aa","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 15:07:26.000000000","message":"Patch Set 21:\n\n* pci-test http://52.27.155.124/pci/319379/21 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"79357ae5cc23f958715bcee540100268073dee81","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 15:56:34.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/21/check/check-dsvm-tempest-vz7-exe-minimal/5a909e2 : FAILURE in 59m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"d259bf901914c95265f16daf4dd416cd95fcf230","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 15:56:36.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/21/check/gate-nova-docs-ubuntu-xenial/9f7c602//doc/build/html/ : SUCCESS in 2m 17s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-nova-pep8-ubuntu-xenial/6e340dd/ : FAILURE in 6m 45s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-nova-python27-db-ubuntu-xenial/5549974/ : FAILURE in 13m 14s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/21/check/gate-nova-python34-db/5949ef7/ : FAILURE in 7m 29s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/21/check/gate-nova-python35-db-nv/4bbeadc/ : FAILURE in 11m 18s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bb5d810/ : FAILURE in 1h 24m 51s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-grenade-dsvm-neutron-ubuntu-xenial/08bcc23/ : FAILURE in 42m 40s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-full-ubuntu-xenial/8adec3b/ : SUCCESS in 50m 56s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-grenade-dsvm-ubuntu-xenial/deb25bb/ : SUCCESS in 41m 21s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-nova-tox-db-functional-ubuntu-xenial/3e594d9/ : SUCCESS in 9m 35s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/21/check/gate-grenade-dsvm-neutron-multinode/e689aec/ : FAILURE in 37m 39s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/79/319379/21/check/gate-grenade-dsvm-multinode-live-migration-nv/3e8bbaa/ : FAILURE in 28m 31s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-cells-ubuntu-xenial/d251552/ : FAILURE in 1h 15m 02s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/74df6a7/ : FAILURE in 43m 18s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f42386f/ : FAILURE in 1h 12m 55s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/de62f93/ : FAILURE in 1h 04m 36s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6aa01f0/ : SUCCESS in 18m 18s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/21/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/0ddc123/ : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f4660c0960fb966437d8116be839c81fdbd92c85","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-11-02 16:01:36.000000000","message":"Patch Set 21:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-neutron-multinode: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n- gate-grenade-dsvm-neutron-ubuntu-xenial: https://bugs.launchpad.net/bugs/1463631\n- gate-nova-pep8-ubuntu-xenial: unrecognized error\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python34-db: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":21},{"id":"ff6309f97ef483db82c354a4fea276160d4d192e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-02 16:16:31.000000000","message":"Patch Set 21:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/21/check/dsvm-tempest-xen/3ae43cc : FAILURE in 1h 23m 13s","accounts_in_message":[],"_revision_number":21},{"id":"3ff228b1100bade3ae970f6cba8748dd8705383f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 16:46:11.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/21/check/dsvm-tempest-kvm/54b92f5 : SUCCESS in 1h 04m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"b741f3b9b94b1de4d794deaa7076ae19242ae271","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-02 18:15:50.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/21/check/dsvm-tempest-nova-network/dadb206 : FAILURE in 1h 20m 36s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/21/check/dsvm-tempest-neutron-network/a76c712 : FAILURE in 1h 42m 58s","accounts_in_message":[],"_revision_number":21},{"id":"0c6a5430de717888b3d0a18637fc977a947dbbd1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-02 18:23:44.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"914a16538a0d9a7fc5b3622e5ea7dcba5e0f4831","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 19:43:16.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/319379/22 : FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"6f23e803c79a11b0355af56898189015ca85c9c3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-02 20:03:56.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"49198c23eda2fc24bcff1a89d30710d2a404b79b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-02 20:36:47.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"2f338161671c8403f93f456fd46a02d3d126d8d2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 20:50:34.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/319379/23 : FAILURE","accounts_in_message":[],"_revision_number":23},{"id":"fbef82bb546e9a303bfe20e0b6d4906c6e4a8d3e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 21:41:48.000000000","message":"Patch Set 24:\n\n* pci-test http://52.27.155.124/pci/319379/24 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"756e6a51d76147aa7e4161e7493bfcc8f1a1a88f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-02 23:24:43.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/24/check/dsvm-tempest-nova-network/0c212e6 : FAILURE in 10m 26s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/24/check/dsvm-tempest-neutron-network/01e0a93 : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":24},{"id":"39a035ffc93319935fc18376d10839c64ed35a0e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 23:30:47.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/24/check/gate-nova-docs-ubuntu-xenial/af2d0bc//doc/build/html/ : SUCCESS in 4m 32s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-nova-pep8-ubuntu-xenial/a13aa90/ : FAILURE in 5m 54s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-nova-python27-db-ubuntu-xenial/dd54a71/ : FAILURE in 16m 59s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/24/check/gate-nova-python34-db/0f8299b/ : FAILURE in 9m 28s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/24/check/gate-nova-python35-db-nv/5a70d4e/ : FAILURE in 15m 56s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/1b893b0/ : FAILURE in 2h 12m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-grenade-dsvm-neutron-ubuntu-xenial/0091ff4/ : FAILURE in 46m 39s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-full-ubuntu-xenial/9deaf13/ : FAILURE in 2h 50m 40s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-grenade-dsvm-ubuntu-xenial/0ea559c/ : FAILURE in 52m 16s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-nova-tox-db-functional-ubuntu-xenial/7c33bb2/ : FAILURE in 15m 21s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/24/check/gate-grenade-dsvm-neutron-multinode/2c819e8/ : FAILURE in 50m 11s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/79/319379/24/check/gate-grenade-dsvm-multinode-live-migration-nv/b1f6fea/ : FAILURE in 40m 24s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-cells-ubuntu-xenial/5c8e3a7/ : FAILURE in 1h 34m 01s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/10e8092/ : FAILURE in 1h 58m 22s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/9042205/ : FAILURE in 2h 31m 43s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/3392972/ : FAILURE in 1h 01m 22s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a24fb8a/ : FAILURE in 24m 58s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/24/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/828a405/ : FAILURE in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"32faf7601763ffb97a6edaf4e76a33552cb0632b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 23:41:21.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/24/check/check-dsvm-tempest-vz7-exe-minimal/e95c47f : FAILURE in 2h 03m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"67d7340a4456aac8c6f654ea0b628320cc77ce62","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 00:04:32.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/24/check/dsvm-tempest-kvm/cba1a0c : FAILURE in 1h 46m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"fca93d63f65354515ac6d378542f2138568968be","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 01:42:43.000000000","message":"Patch Set 24:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/24/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":24},{"id":"52d4114f6ec767d07021ab15f4b227e32c1a0bc6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-03 02:21:39.000000000","message":"Patch Set 24:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/24/check/dsvm-tempest-xen/f580eb2 : FAILURE in 2h 47m 44s","accounts_in_message":[],"_revision_number":24},{"id":"fc41b1ebf6268c98d46a1fb1fc55f8e6f08cfd77","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-03 03:12:55.000000000","message":"Patch Set 21:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":21},{"id":"76ddae3768df897820fcf9493702af088e096eae","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-03 03:30:22.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":24},{"id":"50d8dfdb733ee51c352fe69b3445621432784243","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-03 05:38:58.000000000","message":"Patch Set 24: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/24/1491/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/24/1491/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":24},{"id":"193feb5c997c4c07af8a283d77d375413d15ba7c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-03 14:01:46.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"720f4fffbf66e1b4e6439945ee7e04a86d83f71a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-03 14:49:34.000000000","message":"Patch Set 25:\n\n* pci-test http://52.27.155.124/pci/319379/25 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"5611137cba75d9fb7652e23616192a067b90b9ac","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 15:49:14.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/25/check/dsvm-tempest-kvm/648e4ed : FAILURE in 1h 45m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"1a85ce57141c7dd751b83e2d2d835e8f733829d6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 16:05:38.000000000","message":"Patch Set 25:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/25/check/check-dsvm-tempest-vz7-exe-minimal/29d72c5 : FAILURE in 2h 02m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"af5fa259ed2800c6c8e7157acac2e1b1a8c8ce37","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 16:42:30.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/25/check/gate-nova-docs-ubuntu-xenial/2f8af52//doc/build/html/ : SUCCESS in 3m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-nova-pep8-ubuntu-xenial/d50558c/ : SUCCESS in 5m 01s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-nova-python27-db-ubuntu-xenial/2fbd12b/ : FAILURE in 9m 57s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/25/check/gate-nova-python34-db/1b1ab30/ : FAILURE in 8m 39s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/25/check/gate-nova-python35-db-nv/6bb25ec/ : FAILURE in 8m 12s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/992b732/ : FAILURE in 2h 07m 04s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d5c6902/ : FAILURE in 35m 43s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-grenade-dsvm-ubuntu-xenial/9becc69/ : FAILURE in 1h 24m 49s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-nova-tox-db-functional-ubuntu-xenial/609e0e1/ : FAILURE in 13m 45s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/25/check/gate-grenade-dsvm-neutron-multinode/ad65f34/ : FAILURE in 54m 03s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/25/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a6ba5ed/ : FAILURE in 42m 30s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-cells-ubuntu-xenial/d65f721/ : FAILURE in 1h 54m 59s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7cf1a33/ : FAILURE in 2h 11m 09s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/6ef8943/ : FAILURE in 2h 37m 37s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f87257f/ : FAILURE in 1h 18m 00s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3065256/ : FAILURE in 25m 00s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/25/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/df22b32/ : FAILURE in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ed1463862fa7dfe36a7bf5b3391a8dcb11269382","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-11-03 16:42:59.000000000","message":"Patch Set 25:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-neutron-multinode: unrecognized error\n- gate-grenade-dsvm-neutron-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658 https://bugs.launchpad.net/bugs/1355573\n- gate-grenade-dsvm-ubuntu-xenial: unrecognized error\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python34-db: unrecognized error\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":25},{"id":"1d87fbbf66b60fe0bb4eee18f978b2496126682e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:04:00.000000000","message":"Patch Set 25:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/25/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":25},{"id":"ce4caf2f67ec08739fc54ba372f71c61ad60c535","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-03 18:18:08.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"3cb7d81f1e058fca6a17db7fcc386bb7c273807c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-03 18:37:28.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"63576592b05c2770e679ac833dec0914185f363e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-03 18:37:59.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"119e40d3567ec685ba16ec7eb5ce02ff5947065b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 18:38:02.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"3cf71ae6d3625c660e0a474b462e69066ccf561a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:38:03.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"d6f7552ca2b5034bb80fa7d08d2ea7bae4b47357","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:38:05.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"7ec3c5da14fa7249784b4bbdf46a59213312b3d3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-03 18:38:07.000000000","message":"Patch Set 26:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)\n- check-ibm-tempest-dsvm-neutron-full check-ibm-tempest-dsvm-neutron-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4a50534b56d3c8f108044e7dd4a3c72cfa9c3930","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:38:10.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"a92935a9906efd6a8e34029e94368c04e5c75a81","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:38:12.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":26},{"id":"32e3b8adafbd5e17c6627c0c80fd4da5c0bc271c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-03 18:39:04.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"4f5b384ecc7ac754d1eb513597ef19a7a9ad77cf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-03 19:55:20.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"8b11546aa0815ce42eb96b7979bf1ed8ed933fde","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-03 20:06:24.000000000","message":"Patch Set 26:\n\n* pci-test http://52.27.155.124/pci/319379/26 : FAILURE","accounts_in_message":[],"_revision_number":26},{"id":"fd19fb8ac58d2394f0e280d17bca4d617918dd88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-03 21:25:40.000000000","message":"Patch Set 26:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/26/check/check-ibm-tempest-dsvm-full/bd71204/ : SUCCESS in 57m 29s (non-voting)\n- check-ibm-tempest-dsvm-neutron-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/26/check/check-ibm-tempest-dsvm-neutron-full/c1beba5/ : FAILURE in 1h 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"0b54f36f52fc8a94223aedd75ce3ada3c7672abf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 21:35:20.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/26/check/gate-nova-docs-ubuntu-xenial/f3a0d43//doc/build/html/ : SUCCESS in 3m 32s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-nova-pep8-ubuntu-xenial/bc91193/ : FAILURE in 4m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-nova-python27-db-ubuntu-xenial/54e3c8d/ : FAILURE in 9m 05s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/26/check/gate-nova-python34-db/217a312/ : FAILURE in 8m 58s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/26/check/gate-nova-python35-db-nv/06311b8/ : FAILURE in 7m 20s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e882fbb/ : FAILURE in 1h 33m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-grenade-dsvm-neutron-ubuntu-xenial/bf80f44/ : FAILURE in 31m 45s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-grenade-dsvm-ubuntu-xenial/ece7c69/ : SUCCESS in 25m 25s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-nova-tox-db-functional-ubuntu-xenial/f8c0bb1/ : FAILURE in 9m 48s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/26/check/gate-grenade-dsvm-neutron-multinode/43af9d7/ : FAILURE in 38m 10s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/26/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/437cc84/ : FAILURE in 37m 21s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-cells-ubuntu-xenial/cacd83c/ : FAILURE in 1h 36m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/f73a063/ : FAILURE in 41m 18s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/5cbe191/ : FAILURE in 56m 36s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/9576097/ : FAILURE in 1h 01m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/93a7ab3/ : FAILURE in 23m 07s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/26/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/6b59106/ : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"351d6295058bdcb837fe6070c4133e4827930bf1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 22:19:55.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/26/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":26},{"id":"ec4e9d059f3f47b36e758a04c984df939e317e6c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-04 04:22:22.000000000","message":"Patch Set 26: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/26/1560/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/26/1560/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":26},{"id":"7e3b8113093f53e384f232fd35c0c75e77e5f0c2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-04 06:04:52.000000000","message":"Patch Set 26:\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/26/1562/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/26/1562/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":26},{"id":"5128150b4f4de0d611a9fa9835b38f8de3ea7e96","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-04 09:30:11.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/26/check/tempest-dsvm-full-nfv-xenial/d573d57 : SUCCESS in 1h 39m 13s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/26/check/tempest-dsvm-intel-nfv-xenial/704f746 : FAILURE in 38m 57s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/26/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/685f135 : SUCCESS in 1h 05m 57s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/26/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/4b2a6f0 : FAILURE in 1h 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9b1ab119fa91080e6c6733a98b7663f9fcf0d29b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-07 22:13:03.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"fefa02e48d0699e7a8679e68eece8376ac308c60","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-07 23:18:02.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/27/check/dsvm-tempest-kvm/2814fd1 : FAILURE in 1h 01m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":27},{"id":"082acfb70cd40bbe1d468d200c5464e5ad63a7b0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-07 23:19:23.000000000","message":"Patch Set 27:\n\n* pci-test http://52.27.155.124/pci/319379/27 : FAILURE","accounts_in_message":[],"_revision_number":27},{"id":"891a4f65e9cdd529667de26c7ab339fc706a0ecb","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-07 23:19:52.000000000","message":"Patch Set 27:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/27/check/check-dsvm-tempest-vz7-exe-minimal/be44b21 : FAILURE in 1h 04m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":27},{"id":"e0bf716178e3c04a9ce36ca037a52ff3f6f560af","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-08 00:04:51.000000000","message":"Patch Set 27:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/27/check/dsvm-tempest-xen/3306ac0 : FAILURE in 1h 38m 34s","accounts_in_message":[],"_revision_number":27},{"id":"f8a3d3b7b1dfdfe592b5f610479856a4eae5748b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-08 00:09:16.000000000","message":"Patch Set 27:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/27/check/check-ibm-tempest-dsvm-full-xenial/401214f/ : FAILURE in 1h 18m 49s (non-voting)\n- check-ibm-tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/27/check/check-ibm-tempest-dsvm-neutron-full-xenial/7b7177f/ : FAILURE in 1h 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"31d0eb02ead933f2eb881935d5c38c69d7c54869","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-08 00:25:33.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/27/check/gate-nova-docs-ubuntu-xenial/fe063e9//doc/build/html/ : SUCCESS in 2m 41s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-nova-pep8-ubuntu-xenial/79ce0ce/ : FAILURE in 4m 31s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-nova-python27-db-ubuntu-xenial/b93639d/ : FAILURE in 12m 46s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/27/check/gate-nova-python34-db/e278d07/ : FAILURE in 7m 48s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/27/check/gate-nova-python35-db-nv/3c2c490/ : FAILURE in 8m 07s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/93dae38/ : FAILURE in 1h 21m 10s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c88bfea/ : FAILURE in 52m 15s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-grenade-dsvm-ubuntu-xenial/b05879e/ : SUCCESS in 27m 26s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-nova-tox-db-functional-ubuntu-xenial/af094db/ : FAILURE in 10m 35s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/27/check/gate-grenade-dsvm-neutron-multinode/46e7660/ : FAILURE in 40m 22s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/27/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/d156135/ : FAILURE in 39m 24s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-cells-ubuntu-xenial/b75d9f1/ : FAILURE in 1h 46m 07s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d06ae2a/ : FAILURE in 50m 07s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/6df5686/ : FAILURE in 1h 09m 17s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f593edc/ : FAILURE in 2h 07m 39s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/23aab30/ : FAILURE in 1h 07m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d55e293/ : FAILURE in 23m 54s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/27/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/32de6b4/ : FAILURE in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"5e7847813841c68a0be627ae16d7c2beffc4b14c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-08 00:46:32.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/27/check/dsvm-tempest-neutron-network/dd14392 : FAILURE in 1h 31m 15s","accounts_in_message":[],"_revision_number":27},{"id":"0cbd0765bef11c82e3262548b1654310d7ba83c3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-08 04:37:48.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/27/check/tempest-dsvm-full-nfv-xenial/1bd514a : FAILURE in 1h 51m 16s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/27/check/tempest-dsvm-intel-nfv-xenial/1decb2a : FAILURE in 40m 55s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/27/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/7fd4298 : FAILURE in 1h 20m 02s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/27/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/de6d576 : FAILURE in 2h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"06417df16bd779e00c39ee2e013ab376d4c8288d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-08 05:12:18.000000000","message":"Patch Set 27: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/27/1721/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/27/1721/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":27},{"id":"8da465a76f42384859a0e8fd0d4ce6c2ed917683","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-08 09:52:59.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/27/check-nova/Nova-ML2-Sriov/0748a0f : FAILURE in 1h 00m 28s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/27/check-nova/Nova-MACVTAP-ML2-Sriov/ac162d1 : FAILURE in 58m 08s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":27},{"id":"0ba00ba8011f57501db5f9cbf080a0be264b3cc0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-08 15:57:21.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"af2f455ce6bbace833f3490e72cf865368a6845b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-08 17:30:46.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"24629489bfc8cf39d2d3702ffbe4bc712978aebd","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-08 19:05:19.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/29/check/dsvm-tempest-kvm/8658373 : FAILURE in 1h 01m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"ba45b01363998678923b4edcae132bc916c82ee7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-08 19:12:16.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/29/check/gate-nova-docs-ubuntu-xenial/cac03c6//doc/build/html/ : SUCCESS in 4m 24s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-nova-pep8-ubuntu-xenial/ad0f82b/ : FAILURE in 5m 14s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-nova-python27-db-ubuntu-xenial/018560a/ : FAILURE in 9m 25s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/29/check/gate-nova-python34-db/7ba0118/ : FAILURE in 7m 49s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/29/check/gate-nova-python35-db-nv/30f453b/ : FAILURE in 7m 58s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ba6de03/ : FAILURE in 52m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a1dd2d6/ : SUCCESS in 35m 01s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-grenade-dsvm-ubuntu-xenial/05d758d/ : SUCCESS in 43m 34s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-nova-tox-db-functional-ubuntu-xenial/e84b668/ : FAILURE in 17m 11s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/29/check/gate-grenade-dsvm-neutron-multinode/004909b/ : SUCCESS in 41m 32s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/29/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2bdbcb2/ : FAILURE in 41m 29s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-cells-ubuntu-xenial/bb1c9bc/ : FAILURE in 1h 36m 53s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/af7091b/ : FAILURE in 1h 12m 36s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/88bfdf2/ : FAILURE in 54m 22s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/b49d708/ : FAILURE in 47m 38s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/e07681f/ : FAILURE in 50m 46s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d2ec80e/ : FAILURE in 25m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/29/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/8947371/ : SUCCESS in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"21f6413cb9ca322fb4e7c5b0e148a62256b79727","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-08 19:18:15.000000000","message":"Patch Set 29:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/29/check/dsvm-tempest-xen/97ee45d : FAILURE in 1h 34m 05s","accounts_in_message":[],"_revision_number":29},{"id":"3231fa1003b7e1e133abce3e0553adcc2d518e71","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-08 19:33:25.000000000","message":"Patch Set 29:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/29/check/tempest-dsvm-full-xenial/3cb6b6c/ : FAILURE in 1h 12m 40s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/29/check/tempest-dsvm-neutron-full-xenial/3e7a7a3/ : FAILURE in 1h 59m 16s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"a510d740d7acd7a1d55fb89d5d2e173964768d52","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-08 19:54:36.000000000","message":"Patch Set 28:\n\n* pci-test http://52.27.155.124/pci/319379/28 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"5bd5a9017047d638056b15bad062dd000f808c3c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-08 19:58:23.000000000","message":"Patch Set 29:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/29/check/check-dsvm-tempest-vz7-exe-minimal/ea39371 : FAILURE in 1h 30m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"9cd56b3b930b84e3b762a088744235f5f2fa077a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-08 20:45:43.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/29/check/dsvm-tempest-neutron-network/27fbf4d : FAILURE in 2h 03m 04s","accounts_in_message":[],"_revision_number":29},{"id":"d90ab5b690af392fae19d00ad0d88e633996235e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-08 21:05:44.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"902258b4fde76406b6135a588e662b4a5c2fbcff","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-08 22:06:00.000000000","message":"Patch Set 30:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/30/check/check-dsvm-tempest-vz7-exe-minimal/d4a36f0 : FAILURE in 59m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":30},{"id":"a73e08de0e0359a561b4c11d68701e6d03ab6c9e","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-08 22:07:13.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/30/check/dsvm-tempest-kvm/3f2259a : FAILURE in 1h 00m 30s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":30},{"id":"069e047e7ac15550729e28fdaf20b44b76b57a0b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-08 22:14:22.000000000","message":"Patch Set 29:\n\n* pci-test http://52.27.155.124/pci/319379/29 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"6da07f310ae1d261369c0c12699d323d471aa34f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-08 22:35:45.000000000","message":"Patch Set 30:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/30/check/tempest-dsvm-full-xenial/a4d2cf2/ : FAILURE in 1h 11m 53s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/30/check/tempest-dsvm-neutron-full-xenial/439524b/ : FAILURE in 1h 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"5848da89a1f380442f0c79d58eb8a4691aec4507","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-08 22:40:20.000000000","message":"Patch Set 30:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/30/check/dsvm-tempest-xen/95d0df4 : FAILURE in 1h 32m 16s","accounts_in_message":[],"_revision_number":30},{"id":"a5a0b257ba3bd7fd2dfa5a2aadd35ed665be5381","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-08 23:12:28.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/30/check/dsvm-tempest-neutron-network/d439be5 : FAILURE in 2h 03m 44s","accounts_in_message":[],"_revision_number":30},{"id":"8ad3ab202192b72595368d099b5c8e110c2add6a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-08 23:48:10.000000000","message":"Patch Set 30:\n\n* pci-test http://52.27.155.124/pci/319379/30 : FAILURE","accounts_in_message":[],"_revision_number":30},{"id":"be19dbd5e57c1f08c85bd1e7ca4a13d5b6b39b88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 00:38:04.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/30/check/gate-nova-docs-ubuntu-xenial/c4858ef//doc/build/html/ : SUCCESS in 3m 54s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-nova-pep8-ubuntu-xenial/6be51a3/ : SUCCESS in 4m 45s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-nova-python27-db-ubuntu-xenial/d23a1b1/ : FAILURE in 8m 50s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/30/check/gate-nova-python34-db/d8532ca/ : FAILURE in 9m 54s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/30/check/gate-nova-python35-db-nv/21b1927/ : FAILURE in 14m 08s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8eb1813/ : FAILURE in 46m 30s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8b75046/ : SUCCESS in 57m 04s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-grenade-dsvm-ubuntu-xenial/ff5ba57/ : SUCCESS in 55m 38s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-nova-tox-db-functional-ubuntu-xenial/d43e8d4/ : FAILURE in 10m 22s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/30/check/gate-grenade-dsvm-neutron-multinode/505a11b/ : SUCCESS in 38m 18s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/30/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/3ef782e/ : FAILURE in 42m 03s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-cells-ubuntu-xenial/175a21a/ : FAILURE in 1h 35m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7b502b2/ : FAILURE in 1h 04m 04s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/94bd05a/ : FAILURE in 1h 10m 54s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/5e95640/ : FAILURE in 46m 56s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/9dcd923/ : FAILURE in 46m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/eb6e529/ : FAILURE in 24m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/30/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/475c4fa/ : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2c4c83b8b8c1e03b69e9859d9e97b19e46db2607","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 01:24:00.000000000","message":"Patch Set 30:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/30/ : FAILURE in 4h 12m 51s","accounts_in_message":[],"_revision_number":30},{"id":"897e6c3c67c122b9920c46048319264cb2eb7b6e","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-11-09 10:25:17.000000000","message":"Patch Set 26:\n\nThe main issue we still need to address here is shedule_and_build_instances method for cells v1 scheduler.","accounts_in_message":[],"_revision_number":26},{"id":"3ac0c60196df1a194fa79556c0830e289f4a5998","author":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"date":"2016-11-09 10:29:54.000000000","message":"Patch Set 26:\n\nBy the scheduler I meant cells v1 conductor )","accounts_in_message":[],"_revision_number":26},{"id":"08f258d9eb79c090b131edaa0689ae88735f5d60","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 18:08:08.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/30/check/tempest-dsvm-full-nfv-xenial/ff761de : FAILURE in 2h 08m 31s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/30/check/tempest-dsvm-intel-nfv-xenial/d74795a : SUCCESS in 49m 16s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/30/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/231ef2a : FAILURE in 1h 01m 02s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/30/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/c355f68 : FAILURE in 1h 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"a94ebf862604b6f317d165bd3aba11cb03582f70","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-09 22:53:03.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/30/check-nova/Nova-ML2-Sriov/e7df2d7 : FAILURE in 1h 00m 35s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/30/check-nova/Nova-MACVTAP-ML2-Sriov/adda426 : FAILURE in 1h 04m 22s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":30},{"id":"6b05df96039d1359c0cbcacb6ed03a13ea476228","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-10 16:38:33.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"b412d288a9e22c95068703e0b23a769e122efff7","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-10 17:46:52.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/31/check/dsvm-tempest-kvm/2b09223 : FAILURE in 1h 04m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":31},{"id":"5be3df03e17c913c96db8d43279536671447197c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-10 17:50:59.000000000","message":"Patch Set 31:\n\n* pci-test http://52.27.155.124/pci/319379/31 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"7671f75585d5517155f5f41a203a8402557fbf3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-10 18:06:15.000000000","message":"Patch Set 31:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/31/check/tempest-dsvm-full-xenial/0b31980/ : FAILURE in 1h 15m 50s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/31/check/tempest-dsvm-neutron-full-xenial/2860edd/ : FAILURE in 1h 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"192cd71bb759359dc76c5780367580f3dbc91937","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-10 18:18:45.000000000","message":"Patch Set 31:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/31/check/check-dsvm-tempest-vz7-exe-minimal/a4405bc : FAILURE in 1h 00m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":31},{"id":"b5ff28d628105c4f9fc62a0f5df5d492b1c47989","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-10 18:50:30.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/31/check/gate-nova-docs-ubuntu-xenial/dc82cac//doc/build/html/ : SUCCESS in 4m 04s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-nova-pep8-ubuntu-xenial/6c347e5/ : SUCCESS in 4m 37s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-nova-python27-db-ubuntu-xenial/beecbcf/ : FAILURE in 12m 41s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/31/check/gate-nova-python34-db/ba82b94/ : FAILURE in 8m 14s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/31/check/gate-nova-python35-db-nv/bef2a56/ : FAILURE in 12m 57s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/04a2a2c/ : FAILURE in 51m 03s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-grenade-dsvm-neutron-ubuntu-xenial/55c90bb/ : SUCCESS in 1h 02m 05s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-grenade-dsvm-ubuntu-xenial/c6f5b1b/ : SUCCESS in 54m 33s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-nova-tox-db-functional-ubuntu-xenial/ef5d1db/ : FAILURE in 10m 34s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/31/check/gate-grenade-dsvm-neutron-multinode/ff59e62/ : SUCCESS in 1h 01m 18s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/31/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/646b45b/ : FAILURE in 45m 47s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-cells-ubuntu-xenial/841ac5c/ : FAILURE in 1h 38m 08s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d1bc886/ : FAILURE in 46m 02s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/9e3c16a/ : FAILURE in 1h 12m 36s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/0951de2/ : FAILURE in 53m 36s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/73925a7/ : FAILURE in 1h 04m 19s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/17190c1/ : FAILURE in 26m 50s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/31/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/e856f03/ : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"162d68292c63ba29dc4023441d8c4c9f19399732","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-10 18:58:11.000000000","message":"Patch Set 31:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/31/check/dsvm-tempest-xen/9576c6f : FAILURE in 2h 04m 02s","accounts_in_message":[],"_revision_number":31},{"id":"498059223ec9ed5468562c38edb24843aa85ee2b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-10 20:03:16.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/31/check/dsvm-tempest-neutron-network/b2818ae : FAILURE in 2h 15m 57s","accounts_in_message":[],"_revision_number":31},{"id":"9487144cba1176f5703811337d6d9bef59fa7043","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-10 21:31:12.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":31},{"id":"79dafd433898470b7554335e71c1f6119fc2ec7e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-10 21:57:56.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"e59ce8d32381d8a1569a77539b3315650a62afe4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-10 22:24:01.000000000","message":"Patch Set 32:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/32/check/tempest-dsvm-full-xenial/06adf54/ : FAILURE in 23m 47s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/32/check/tempest-dsvm-neutron-full-xenial/088a232/ : FAILURE in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"3a2236a92479a484618352c72d7455c134f67ed8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-10 22:26:50.000000000","message":"Patch Set 32:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/32/check/check-dsvm-tempest-vz7-exe-minimal/41f1445 : FAILURE in 26m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":32},{"id":"b256ad15bda80e11d74ed5ab68579903c19372fd","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-10 22:28:07.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/32/check/dsvm-tempest-kvm/bddd298 : FAILURE in 26m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":32},{"id":"d7e18097c98e3dc35cf5a04598df6cd2c472b148","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-10 22:37:54.000000000","message":"Patch Set 32:\n\n* pci-test http://52.27.155.124/pci/319379/32 : FAILURE","accounts_in_message":[],"_revision_number":32},{"id":"f3c665a225fa836e3478dc459b0437bf4510189f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-10 22:40:56.000000000","message":"Patch Set 32:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/32/check/dsvm-tempest-xen/7c345ee : FAILURE in 28m 26s","accounts_in_message":[],"_revision_number":32},{"id":"166c3343dd70dc65dc068b5ee23e578acf916aef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-10 23:04:57.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/32/check/gate-nova-docs-ubuntu-xenial/f9022c2//doc/build/html/ : SUCCESS in 3m 19s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-nova-pep8-ubuntu-xenial/e9f9f92/ : SUCCESS in 4m 35s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-nova-python27-db-ubuntu-xenial/e0153fc/ : FAILURE in 9m 55s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/32/check/gate-nova-python34-db/232d337/ : FAILURE in 38m 55s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/32/check/gate-nova-python35-db-nv/f066675/ : FAILURE in 7m 52s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/1c2fc25/ : FAILURE in 12m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-grenade-dsvm-neutron-ubuntu-xenial/f4a64e5/ : FAILURE in 25m 14s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-grenade-dsvm-ubuntu-xenial/565c3a4/ : FAILURE in 24m 21s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-nova-tox-db-functional-ubuntu-xenial/b4078e1/ : FAILURE in 10m 40s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/79/319379/32/check/gate-grenade-dsvm-neutron-multinode/aeb32c2/ : FAILURE in 32m 17s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/32/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/6f68949/ : FAILURE in 29m 23s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-cells-ubuntu-xenial/1c88a6e/ : FAILURE in 10m 59s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ea294d9/ : FAILURE in 12m 58s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/c2dcd02/ : FAILURE in 24m 45s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/ece677c/ : FAILURE in 13m 42s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/915e253/ : FAILURE in 12m 17s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/04297a6/ : FAILURE in 14m 43s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/79/319379/32/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/1fecdb6/ : FAILURE in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"a81105124f08ee2551e1f63911663bd9c913859b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-10 23:14:53.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/32/check/dsvm-tempest-neutron-network/538bd42 : FAILURE in 36m 45s","accounts_in_message":[],"_revision_number":32},{"id":"af39c43151ba8fff8f6cb0af059da0add729bdf0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-10 23:23:39.000000000","message":"Patch Set 32: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/32/1889/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/32/1889/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":32},{"id":"f01ce381c9f715319873bcd13317e3884d019346","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-11 05:12:22.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/32/check/tempest-dsvm-full-nfv-xenial/dd78f94 : FAILURE in 29m 55s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/32/check/tempest-dsvm-intel-nfv-xenial/4b98261 : FAILURE in 29m 18s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/32/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/3868379 : FAILURE in 50m 06s","accounts_in_message":[],"_revision_number":32},{"id":"4863362f552497edc664fadae0515012bc5b8468","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-11 21:01:18.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"c841a93363df2bb1192b663ab335912640ca2be8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-11 21:30:47.000000000","message":"Patch Set 33:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/33/check/tempest-dsvm-full-xenial/3c0960d/ : FAILURE in 22m 42s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/33/check/tempest-dsvm-neutron-full-xenial/816031f/ : FAILURE in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"31b43461a57116a380f3931f91c077f10c18b1e9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-11 21:32:14.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/33/check/dsvm-tempest-kvm/2771090 : FAILURE in 26m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":33},{"id":"0bc8892e410cd8eecb2ac61e2dce9a7391729248","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-11 21:40:31.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/33/check/gate-nova-docs-ubuntu-xenial/dc829b3//doc/build/html/ : SUCCESS in 2m 41s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-nova-pep8-ubuntu-xenial/d820c24/ : FAILURE in 4m 38s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-nova-python27-db-ubuntu-xenial/795e52d/ : FAILURE in 9m 26s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/33/check/gate-nova-python34-db/20638e7/ : FAILURE in 7m 38s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/33/check/gate-nova-python35-db-nv/2b810ee/ : FAILURE in 8m 10s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f8c93d5/ : FAILURE in 12m 32s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-grenade-dsvm-neutron-ubuntu-xenial/0121460/ : FAILURE in 26m 59s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-grenade-dsvm-ubuntu-xenial/4748e21/ : FAILURE in 21m 15s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-nova-tox-db-functional-ubuntu-xenial/3950105/ : FAILURE in 16m 08s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/33/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/6018711/ : FAILURE in 28m 24s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-cells-ubuntu-xenial/9564b43/ : FAILURE in 23m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a3b6147/ : FAILURE in 12m 25s\n- gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-neutron-linuxbridge-ubuntu-xenial/471eedf/ : FAILURE in 24m 07s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f10d238/ : FAILURE in 15m 00s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f4ffee5/ : FAILURE in 24m 05s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/26bdd7e/ : FAILURE in 13m 31s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/33/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/2b3c386/ : FAILURE in 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"fb20c84f5c219b58d26785fc999c90aef6232e88","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-11 21:46:49.000000000","message":"Patch Set 33:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/33/check/dsvm-tempest-xen/92a801b : FAILURE in 26m 34s","accounts_in_message":[],"_revision_number":33},{"id":"c1adc295c4c58a35dbbfd143153b3a6825762689","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-11 21:47:47.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/33/check/dsvm-tempest-neutron-network/45de324 : FAILURE in 30m 20s","accounts_in_message":[],"_revision_number":33},{"id":"df6793e8f0be909b45d8d9a8a52c9a5eea4e7558","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-11 22:24:49.000000000","message":"Patch Set 33:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/33/check/check-dsvm-tempest-vz7-exe-minimal/e91b4ff : FAILURE in 33m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":33},{"id":"6ddbe14c6d50b11021683f7bc7f2c84f9bfba52f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-11 22:42:48.000000000","message":"Patch Set 33: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/319379/33/1973/ : FAILURE\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/319379/33/1973/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":33},{"id":"5473e1d6adfe6f7d02e6dcb7d7cdd5eb461c339e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-11 23:17:58.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/33/check/tempest-dsvm-full-nfv-xenial/b230027 : FAILURE in 44s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/33/check/tempest-dsvm-intel-nfv-xenial/568d47b : FAILURE in 28m 35s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/33/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9028b55 : FAILURE in 49m 36s","accounts_in_message":[],"_revision_number":33},{"id":"fe5496c9dab0d7cf9f0736f01e53a942220325b9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-11 23:59:43.000000000","message":"Patch Set 33:\n\n* pci-test http://52.27.155.124/pci/319379/33 : FAILURE","accounts_in_message":[],"_revision_number":33},{"id":"6eafb4e6effd3f9464da4489aeaef1745b29d460","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-12 09:42:20.000000000","message":"Patch Set 33:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/33/ : FAILURE in 38m 11s","accounts_in_message":[],"_revision_number":33},{"id":"3a04b530f7da7e9336f29cf9c272ed98ef8b4d16","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-13 08:46:39.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/33/check-nova/Nova-ML2-Sriov/19a8a68 : FAILURE in 49m 49s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/33/check-nova/Nova-MACVTAP-ML2-Sriov/0e3b71a : FAILURE in 52m 49s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":33},{"id":"8d21935e746113cf938424d13cd8e37a19ec238a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-30 20:17:18.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"3e4d3cd8c5c74f533e5a13d232fef55856ea6567","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-30 21:07:17.000000000","message":"Patch Set 34:\n\n* pci-test http://52.27.155.124/pci/319379/34 : FAILURE","accounts_in_message":[],"_revision_number":34},{"id":"97b4a5240bbac56bb7ba2d1182a4a177199cb26a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-30 21:23:39.000000000","message":"Patch Set 34:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/34/check/check-dsvm-tempest-vz7-exe-minimal/ddcd8f4 : FAILURE in 1h 04m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"602fbafd6e32cd58c5badb6e207c099d1794fb8a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-30 21:39:30.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/34/check/dsvm-tempest-kvm/549cc83 : FAILURE in 1h 19m 39s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/34/check/dsvm-tempest-exe/9badeda : FAILURE in 43m 22s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"4457c24b7f1800235053fc7797e7ef8070d064b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-30 22:14:07.000000000","message":"Patch Set 34:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/34/check/tempest-dsvm-full-xenial/6d707bb/ : FAILURE in 1h 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"91e6e7407121052bbd02822742c7343aa27aa58c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-30 23:17:08.000000000","message":"Patch Set 34:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/34/check/dsvm-tempest-xen/e7a480a : FAILURE in 2h 48m 44s","accounts_in_message":[],"_revision_number":34},{"id":"4bed648dc503c92a08d62ffad026e7b4207b27bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-30 23:20:20.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/34/check/gate-nova-docs-ubuntu-xenial/df9592b//doc/build/html/ : SUCCESS in 3m 55s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-nova-pep8-ubuntu-xenial/ce98934/ : FAILURE in 4m 45s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-nova-python27-db-ubuntu-xenial/a6b44b9/ : FAILURE in 16m 35s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/34/check/gate-nova-python34-db/6f9f501/ : FAILURE in 7m 58s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/34/check/gate-nova-python35-db-nv/83b23e1/ : FAILURE in 8m 48s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6554398/ : FAILURE in 1h 13m 07s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-grenade-dsvm-neutron-ubuntu-xenial/38a33f9/ : FAILURE in 37m 17s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-nova-tox-db-functional-ubuntu-xenial/3d5adef/ : FAILURE in 4m 42s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c995fb8/ : FAILURE in 42m 47s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/34/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/37d0963/ : FAILURE in 2h 49m 53s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-cells-ubuntu-xenial/f5bfa58/ : FAILURE in 1h 44m 38s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7b12782/ : FAILURE in 1h 39m 33s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f93789d/ : FAILURE in 1h 01m 49s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/76b4b28/ : FAILURE in 1h 28m 26s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/cea9417/ : FAILURE in 37m 11s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/34/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/6c21f45/ : SUCCESS in 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"fd7009c2bf9b4c1bb0d537b131d372a4ae4d0906","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-30 23:27:39.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/34/check/dsvm-tempest-neutron-network/6c0ee88 : FAILURE in 2h 41m 39s","accounts_in_message":[],"_revision_number":34},{"id":"7468652face466ac5838db5b1905172364a6128f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-11-30 23:41:15.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"924f5e76b41eebd45e89bc2b228a864e0af39978","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-30 23:50:53.000000000","message":"Patch Set 35:\n\n* pci-test http://52.27.155.124/pci/319379/35 : FAILURE","accounts_in_message":[],"_revision_number":35},{"id":"dffe79b4aea77839e8bdfbc3d42b33e5539b8e8e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-01 00:50:30.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/35/check/check-dsvm-tempest-vz7-exe-minimal/cb5930d : FAILURE in 1h 07m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"ff78e866b49daf26e9875e882031cd60f325fdc9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-01 01:17:41.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/35/check/gate-nova-docs-ubuntu-xenial/8ce4fdc//doc/build/html/ : SUCCESS in 3m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-nova-pep8-ubuntu-xenial/36a62cf/ : FAILURE in 4m 49s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-nova-python27-db-ubuntu-xenial/528b649/ : FAILURE in 9m 19s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/35/check/gate-nova-python34-db/9d6ef5a/ : FAILURE in 13m 30s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/35/check/gate-nova-python35-db-nv/6998f4b/ : FAILURE in 8m 04s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5d4be6a/ : FAILURE in 1h 11m 01s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b09b4b1/ : FAILURE in 35m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-nova-tox-db-functional-ubuntu-xenial/a1aedeb/ : FAILURE in 45m 19s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8618850/ : FAILURE in 39m 14s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/35/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/647a0b0/ : FAILURE in 38m 51s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-cells-ubuntu-xenial/ecf8699/ : FAILURE in 1h 29m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/844e1f0/ : FAILURE in 1h 17m 50s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/337ee88/ : FAILURE in 1h 03m 37s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/507f775/ : FAILURE in 1h 32m 43s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8fc3adb/ : FAILURE in 24m 14s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/35/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/d314140/ : SUCCESS in 25m 48s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"97f3b5f32a4275c88763376dc88d4088cd0bb6f8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-01 01:25:40.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/35/check/dsvm-tempest-kvm/d2a6bd8 : FAILURE in 1h 03m 18s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/35/check/dsvm-tempest-exe/9682a64 : FAILURE in 38m 15s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"e3b4adf6a74ec766601fc4111e7bfa4915eabcfd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-01 01:36:33.000000000","message":"Patch Set 35:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/35/check/tempest-dsvm-full-xenial/f5db942/ : FAILURE in 1h 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"15ec5abaeaf39b70f5230ad86918afc30f4eeed8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-01 02:18:36.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/35/check/dsvm-tempest-neutron-network/23957b9 : FAILURE in 2h 30m 53s","accounts_in_message":[],"_revision_number":35},{"id":"70fcb7a3cbd638febf560439900f417576f57500","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-01 02:39:50.000000000","message":"Patch Set 35:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/35/check/dsvm-tempest-xen/b14ca83 : FAILURE in 2h 48m 08s","accounts_in_message":[],"_revision_number":35},{"id":"af76c114dae09c27d26f11b451c803ad48def93a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-01 09:36:40.000000000","message":"Patch Set 35:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/35/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":35},{"id":"eede7e4137fd167ead0c54aabfe14b0c6fc9c22f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-12-01 12:10:03.000000000","message":"Patch Set 35: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":35},{"id":"68aa7d028ca4921c64d03d4445244d12f660d441","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-01 12:10:16.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":35},{"id":"afa6beb08f66504fb6e0366129d6e381ce820a61","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-01 12:10:32.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":35},{"id":"0007dc80608659ff38ae9ff6eb8bc5a1c17e538c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-12-01 13:53:36.000000000","message":"Patch Set 35:\n\n- ext-nova-dsvm http://207.189.188.190/logs/ext-nova-dsvm/319379/35/2760/ : FAILURE\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/35/2760/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":35},{"id":"c1a25ce05e717ee81e44add916b62e9cb0aa385e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-02 00:25:31.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"83388ba96141a6310873a48e9de4e5017b56d878","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-02 01:07:19.000000000","message":"Patch Set 36:\n\n* pci-test http://52.27.155.124/pci/319379/36 : FAILURE","accounts_in_message":[],"_revision_number":36},{"id":"4c2b37af57ea23a9226e8b2773f97eb710329f89","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-02 01:30:21.000000000","message":"Patch Set 36:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/36/check/check-dsvm-tempest-vz7-exe-minimal/14630db : FAILURE in 1h 02m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":36},{"id":"0bee53bb6e07d7aac2aae4140337938a07591a79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-02 02:13:01.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/36/check/gate-nova-docs-ubuntu-xenial/fe5a32f//doc/build/html/ : SUCCESS in 2m 38s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-nova-pep8-ubuntu-xenial/4f9dc3d/ : FAILURE in 3m 56s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-nova-python27-db-ubuntu-xenial/9a17286/ : FAILURE in 8m 48s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/36/check/gate-nova-python34-db/b377ca1/ : FAILURE in 8m 14s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/36/check/gate-nova-python35-db-nv/5f76ea1/ : FAILURE in 14m 02s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c450da1/ : FAILURE in 1h 11m 55s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3b04264/ : FAILURE in 58m 35s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-nova-tox-db-functional-ubuntu-xenial/f2f6644/ : FAILURE in 15m 26s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/89a53af/ : FAILURE in 43m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/36/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/0d7a505/ : FAILURE in 42m 38s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-cells-ubuntu-xenial/945ceb5/ : FAILURE in 1h 43m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/2f880d8/ : FAILURE in 1h 33m 53s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f5b11cf/ : FAILURE in 1h 03m 18s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/953bd9c/ : FAILURE in 1h 11m 17s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/39a3933/ : FAILURE in 24m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/36/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/5a26940/ : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"3d4016638b24c1f06f9f3cc11d4c61404c9c3ebf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-02 02:15:10.000000000","message":"Patch Set 36:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/36/check/tempest-dsvm-full-xenial/caad62e/ : FAILURE in 1h 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"9ef7573ace6b6b142616d78039ecd5936c3fa000","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-02 03:16:36.000000000","message":"Patch Set 36:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/36/check/dsvm-tempest-xen/5f52344 : FAILURE in 2h 38m 21s","accounts_in_message":[],"_revision_number":36},{"id":"bd5e6601baf491da2a913239b4bd247481e5ad0e","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-02 03:29:38.000000000","message":"Patch Set 36:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/36/check/dsvm-tempest-kvm/160c7c3 : FAILURE in 1h 12m 40s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/36/check/dsvm-tempest-exe/1e768bf : FAILURE in 37m 00s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":36},{"id":"864fef97a47117e6ec7efd83b414fdd0dc0cc560","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-02 04:09:15.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/36/check/dsvm-tempest-neutron-network/f3f5dfb : FAILURE in 2h 23m 37s","accounts_in_message":[],"_revision_number":36},{"id":"aaf0797f212092b3c8301b8a87d2a6b3f49a936f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-02 04:28:49.000000000","message":"Patch Set 36:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/36/ : FAILURE in 4h 00m 01s","accounts_in_message":[],"_revision_number":36},{"id":"fd00f496ce05211871adb706d6ec3991536794b0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-02 17:27:53.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"6bbf27d5bab863fd496ee1e6df666d10208c594e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-02 18:36:53.000000000","message":"Patch Set 37:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/37/check/tempest-dsvm-full-xenial/92fded9/ : SUCCESS in 1h 06m 22s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"05a9c4be988c3069a8f7d4803b41fe2865a3f36c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-02 18:56:27.000000000","message":"Patch Set 37:\n\n* pci-test http://52.27.155.124/pci/319379/37 : SUCCESS","accounts_in_message":[],"_revision_number":37},{"id":"e02c88de13e1e5c049d211181a087228c1e5c061","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-02 19:09:55.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/37/check/gate-nova-docs-ubuntu-xenial/bc7e594//doc/build/html/ : SUCCESS in 3m 34s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-nova-pep8-ubuntu-xenial/b5dbc4d/ : SUCCESS in 4m 04s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-nova-python27-db-ubuntu-xenial/c05a804/ : FAILURE in 7m 51s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/37/check/gate-nova-python34-db/93a1d70/ : FAILURE in 10m 26s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/37/check/gate-nova-python35-db-nv/ed5d90a/ : FAILURE in 9m 35s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3ffa04a/ : SUCCESS in 1h 26m 46s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7d3013b/ : SUCCESS in 35m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-nova-tox-db-functional-ubuntu-xenial/a1c3a82/ : FAILURE in 14m 44s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a27fc87/ : SUCCESS in 35m 59s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/37/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/afef418/ : FAILURE in 1h 05m 12s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-cells-ubuntu-xenial/66308c9/ : FAILURE in 1h 36m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/802750c/ : SUCCESS in 54m 38s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/e9a3a55/ : FAILURE in 48m 22s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/20b0734/ : SUCCESS in 1h 18m 47s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/4a2e33f/ : FAILURE in 22m 16s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/37/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8263b26/ : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"ca058c7778f893d84877f712c789452fcc85d1b9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-02 20:20:01.000000000","message":"Patch Set 37:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/37/ : FAILURE in 1h 07m 24s","accounts_in_message":[],"_revision_number":37},{"id":"38da1e808cc68f64572465bf2c0f4689f1d18203","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-02 21:36:47.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/37/check/dsvm-tempest-neutron-network/d5becf0 : FAILURE in 2h 05m 37s","accounts_in_message":[],"_revision_number":37},{"id":"ef5a077cd5d72a90cc598b2bd0948564f0e30a36","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-02 21:41:14.000000000","message":"Patch Set 37:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/37/check/dsvm-tempest-xen/9601fbb : FAILURE in 2h 26m 14s","accounts_in_message":[],"_revision_number":37},{"id":"ff055b7f6a224f869a2c2d536a07598ed572a52d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-02 22:26:38.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"121093ac913e72da61798e0cd6bbc17167bde4e8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-02 23:19:35.000000000","message":"Patch Set 38:\n\n* pci-test http://52.27.155.124/pci/319379/38 : SUCCESS","accounts_in_message":[],"_revision_number":38},{"id":"13066811736e79f538b01d215433d1af6eaf0563","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-02 23:29:59.000000000","message":"Patch Set 38:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/38/check/tempest-dsvm-full-xenial/52dd810/ : SUCCESS in 58m 49s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"f84c24233770d4dc29d474ffd69f4439b83c3da3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-02 23:31:42.000000000","message":"Patch Set 38:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/38/ : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":38},{"id":"f55fb41e29828d0f0d41e10dfd8c46538965042c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-03 00:06:34.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/38/check/gate-nova-docs-ubuntu-xenial/4f222c2//doc/build/html/ : SUCCESS in 3m 59s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-nova-pep8-ubuntu-xenial/7689b01/ : FAILURE in 5m 37s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-nova-python27-db-ubuntu-xenial/a03c1a3/ : FAILURE in 7m 34s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/38/check/gate-nova-python34-db/53f32e7/ : FAILURE in 7m 18s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/38/check/gate-nova-python35-db-nv/906cd80/ : FAILURE in 6m 54s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a5d5a04/ : SUCCESS in 1h 07m 18s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-grenade-dsvm-neutron-ubuntu-xenial/839efea/ : SUCCESS in 32m 12s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-nova-tox-db-functional-ubuntu-xenial/74e8625/ : FAILURE in 6m 54s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b7b2245/ : SUCCESS in 38m 10s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/38/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2310874/ : FAILURE in 39m 07s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-cells-ubuntu-xenial/89e33dc/ : FAILURE in 1h 33m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e78c292/ : SUCCESS in 57m 07s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/d9c550c/ : FAILURE in 1h 05m 00s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/220a218/ : FAILURE in 49m 12s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/be653ab/ : FAILURE in 21m 35s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/38/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/34df1ab/ : SUCCESS in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"9bc4e3cdff18d6835431041f77d8eaec88b388a1","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-03 00:53:33.000000000","message":"Patch Set 38:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/38/check/dsvm-tempest-xen/ee426e4 : SUCCESS in 2h 07m 32s","accounts_in_message":[],"_revision_number":38},{"id":"cc6d96042f98abe8af2a99c2c97398d46e6c873d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-03 01:20:54.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"7da49c2c8809492fb40a974258e622c7a717977f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-03 01:47:24.000000000","message":"Patch Set 39:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":39},{"id":"5d9a60568942a8a35ae407ad3d3bd3a04fe764ed","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-03 02:09:52.000000000","message":"Patch Set 39:\n\n* pci-test http://52.27.155.124/pci/319379/39 : SUCCESS","accounts_in_message":[],"_revision_number":39},{"id":"43c1b8f5f8f7f100f5875df553bfe7fdf18727c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-03 02:27:36.000000000","message":"Patch Set 39:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/39/check/tempest-dsvm-full-xenial/9ba78b5/ : SUCCESS in 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"3564d461e82b7308a9aac515cf03ec214d80b73e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-03 03:15:15.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/39/check/gate-nova-docs-ubuntu-xenial/e049b2b//doc/build/html/ : SUCCESS in 2m 43s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-nova-pep8-ubuntu-xenial/15501d3/ : SUCCESS in 3m 52s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-nova-python27-db-ubuntu-xenial/5045456/ : FAILURE in 7m 18s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/39/check/gate-nova-python34-db/cc92354/ : FAILURE in 7m 11s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/39/check/gate-nova-python35-db-nv/0dd86b9/ : FAILURE in 7m 02s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/23ea5ee/ : SUCCESS in 51m 09s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-grenade-dsvm-neutron-ubuntu-xenial/133258c/ : SUCCESS in 54m 49s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-nova-tox-db-functional-ubuntu-xenial/281fc41/ : FAILURE in 10m 47s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/df97695/ : SUCCESS in 37m 43s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/39/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/357b2dc/ : FAILURE in 40m 10s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-cells-ubuntu-xenial/bb44ea7/ : FAILURE in 1h 45m 24s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b188150/ : FAILURE in 54m 31s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/c0aac79/ : FAILURE in 1h 13m 03s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/5311dd4/ : FAILURE in 1h 12m 21s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c3a1b6f/ : FAILURE in 41m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/39/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c1aa4b6/ : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"085e2e8422a07bb49b88fbbf315eb56b203f986d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-03 03:51:32.000000000","message":"Patch Set 39:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/39/check/dsvm-tempest-xen/d4c2a03 : FAILURE in 2h 07m 48s","accounts_in_message":[],"_revision_number":39},{"id":"ea0e49d5c026046f87a46e05a502208d3dc7834e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-03 04:21:57.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/39/check/dsvm-tempest-neutron-network/cb68b1b : SUCCESS in 2h 32m 54s","accounts_in_message":[],"_revision_number":39},{"id":"39bc2dcffb30a8cceac9a365ea17cadebff1c5db","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-03 04:45:24.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/39/check/dsvm-tempest-kvm/55c2637 : SUCCESS in 1h 03m 25s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/39/check/dsvm-tempest-exe/d531d19 : FAILURE in 45m 36s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":39},{"id":"b66ff7158bd74c627fd512df6eb7fd7608022550","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-03 12:53:51.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/39/check-nova/Nova-ML2-Sriov/b5757f7 : FAILURE in 54m 51s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/39/check-nova/Nova-MACVTAP-ML2-Sriov/080c4e2 : FAILURE in 56m 10s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":39},{"id":"acfb8225413a34582d28efa06462c302377e5402","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-04 19:53:43.000000000","message":"Patch Set 39:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/39/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/db0997f : FAILURE in 1h 14m 52s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":39},{"id":"de1ce62cb31566e47e571e698e799b2aba32ba7b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-05 16:29:32.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"994e00b4ec9e10fdbc311b89a9f813fe2f8fb14c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-05 17:38:12.000000000","message":"Patch Set 40:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/40/check/tempest-dsvm-full-xenial/d525479/ : SUCCESS in 1h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"f870a5233240d5aa1dd38802334c3cc2b1392b93","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-05 18:18:43.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/40/check/gate-nova-docs-ubuntu-xenial/7f9ac2b//doc/build/html/ : SUCCESS in 5m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-nova-pep8-ubuntu-xenial/02b79e8/ : SUCCESS in 8m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-nova-python27-db-ubuntu-xenial/25024ec/ : FAILURE in 12m 53s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/40/check/gate-nova-python34-db/3d49e7b/ : FAILURE in 10m 07s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/40/check/gate-nova-python35-db-nv/637cdca/ : FAILURE in 10m 19s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f5d3686/ : FAILURE in 1h 12m 20s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-grenade-dsvm-neutron-ubuntu-xenial/f1aa23c/ : SUCCESS in 41m 47s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-nova-tox-db-functional-ubuntu-xenial/38373e1/ : FAILURE in 12m 01s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c8a4875/ : SUCCESS in 42m 11s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/40/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/c4f86ad/ : FAILURE in 47m 24s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-cells-ubuntu-xenial/fde81c4/ : FAILURE in 1h 38m 26s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/54da587/ : SUCCESS in 58m 20s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/178aabc/ : FAILURE in 55m 16s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/271eb62/ : SUCCESS in 56m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5aebfee/ : FAILURE in 25m 56s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/40/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/de42c25/ : SUCCESS in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"d29841cf3c3db1babf083905ec0ca674e9ccce2b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-05 18:59:46.000000000","message":"Patch Set 40:\n\n* pci-test http://52.27.155.124/pci/319379/40 : SUCCESS","accounts_in_message":[],"_revision_number":40},{"id":"d554c8c1b99617c61869e977dda6c8aea5493942","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-05 19:05:26.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/40/check/check-dsvm-tempest-vz7-exe-minimal/3648a2a : SUCCESS in 1h 03m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":40},{"id":"74f2d354cd349480ee54a07c068a9afb9668bfa1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-05 19:56:37.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/40/check/dsvm-tempest-neutron-network/f2c558d : FAILURE in 2h 12m 46s","accounts_in_message":[],"_revision_number":40},{"id":"c1c2749388092676974a8fb9838bc10adb9ebb75","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-05 20:20:26.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/40/check/dsvm-tempest-xen/8b1b22e : SUCCESS in 2h 14m 30s","accounts_in_message":[],"_revision_number":40},{"id":"5b0f87ad82439084f7310847b8162faf5bb86c17","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-05 23:26:35.000000000","message":"Patch Set 40:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/40/check/dsvm-tempest-kvm/c146017 : SUCCESS in 59m 35s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/40/check/dsvm-tempest-exe/492bba5 : FAILURE in 45m 09s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":40},{"id":"f94bd4725006d54d5fd831c37b9ecfed3e3607c6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-05 23:42:11.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"ae66495bb7d9216c75af7469194fcba8cd576709","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-06 00:39:39.000000000","message":"Patch Set 41:\n\n* pci-test http://52.27.155.124/pci/319379/41 : SUCCESS","accounts_in_message":[],"_revision_number":41},{"id":"0b1855eb0ce6878e35db31ea333f5c9b7a1afb2d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-06 01:23:08.000000000","message":"Patch Set 41:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/41/check/tempest-dsvm-full-xenial/8124340/ : SUCCESS in 1h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"dd4fa16eadaad17363f65194f4095298aa6f5328","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-06 01:28:25.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/41/check/gate-nova-docs-ubuntu-xenial/c30576e//doc/build/html/ : SUCCESS in 2m 43s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-nova-pep8-ubuntu-xenial/42a1734/ : SUCCESS in 7m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-nova-python27-db-ubuntu-xenial/b6afb85/ : SUCCESS in 7m 41s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/41/check/gate-nova-python34-db/75e4193/ : SUCCESS in 12m 07s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/41/check/gate-nova-python35-db-nv/5229671/ : SUCCESS in 7m 32s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3682ca6/ : FAILURE in 1h 16m 12s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d6bda90/ : SUCCESS in 48m 32s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-nova-tox-db-functional-ubuntu-xenial/a7b7c0b/ : FAILURE in 10m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e5c84fe/ : SUCCESS in 35m 15s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/41/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/337db27/ : FAILURE in 39m 56s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-cells-ubuntu-xenial/0f96f17/ : FAILURE in 1h 36m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5b69d95/ : SUCCESS in 57m 15s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/66b8137/ : FAILURE in 1h 20m 44s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/00e429d/ : SUCCESS in 48m 58s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a95ef09/ : FAILURE in 1h 01m 09s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/41/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/a15e03d/ : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"7d34ad9f56c08dfd36013a6afef0b6c77157d0d9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-06 02:24:13.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/41/check/dsvm-tempest-xen/573f428 : SUCCESS in 2h 18m 01s","accounts_in_message":[],"_revision_number":41},{"id":"8f1cc20ce2987a41ca8b3f095d5f2241a2c6e380","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-06 03:28:39.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/41/check/check-dsvm-tempest-vz7-exe-minimal/76603b7 : SUCCESS in 57m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":41},{"id":"2dc1593f5e0f66142b3fa0804b104c8397a0793a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-06 03:39:33.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/41/check/dsvm-tempest-neutron-network/d2293be : SUCCESS in 2h 28m 23s","accounts_in_message":[],"_revision_number":41},{"id":"4e090c6da4c52f0953646accec39eb23a140fd43","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-06 12:20:46.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/41/ : SUCCESS in 1h 53m 53s","accounts_in_message":[],"_revision_number":41},{"id":"5fe9adf981f1388f1d8127da7f156f19e11583d6","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-06 14:01:23.000000000","message":"Patch Set 41:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/41/check/dsvm-tempest-kvm/a3c33fc : SUCCESS in 58m 46s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/41/check/dsvm-tempest-exe/31732f1 : FAILURE in 42m 53s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":41},{"id":"50b21c1f906d79d9001dcb61f191490cbdf40389","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-06 17:20:51.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"0efb641969b5618558a3d3bb3d8021c607892f58","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-06 18:17:49.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"b9a7bbc0c09e656ab170bf866be663928939eeaa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-06 19:47:47.000000000","message":"Patch Set 42:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/42/check/tempest-dsvm-full-xenial/0c9383a/ : FAILURE in 1h 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"67cde83cf363d543310df0acb7dfdc717e9e0111","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-06 20:25:04.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/42/check/gate-nova-docs-ubuntu-xenial/353244c//doc/build/html/ : SUCCESS in 5m 05s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-nova-pep8-ubuntu-xenial/563abfb/ : SUCCESS in 4m 03s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-nova-python27-db-ubuntu-xenial/7dd19bc/ : SUCCESS in 8m 01s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/42/check/gate-nova-python34-db/b83e0b2/ : SUCCESS in 11m 56s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/42/check/gate-nova-python35-db-nv/059f013/ : SUCCESS in 8m 13s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/dd6d3c4/ : FAILURE in 49m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-grenade-dsvm-neutron-ubuntu-xenial/37b476d/ : SUCCESS in 34m 46s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-nova-tox-db-functional-ubuntu-xenial/45af8ac/ : FAILURE in 45m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ef0d46c/ : SUCCESS in 39m 12s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/42/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/92b69f7/ : FAILURE in 40m 07s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-cells-ubuntu-xenial/624e94a/ : FAILURE in 1h 56m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4a1be60/ : SUCCESS in 1h 21m 58s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/3cb0b16/ : FAILURE in 49m 42s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/391ddda/ : SUCCESS in 1h 16m 34s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a343a8f/ : FAILURE in 40m 10s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/42/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/9b1f6dd/ : SUCCESS in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"5f9eebe2333128641e6f028fadd75bd45f4b2dee","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-06 21:29:12.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/42/check/dsvm-tempest-xen/249ecb1 : SUCCESS in 2h 23m 20s","accounts_in_message":[],"_revision_number":42},{"id":"86cd72ab1c3256c1c45ebaf70fbe3667aea6d813","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-06 21:43:57.000000000","message":"Patch Set 42:\n\n* pci-test http://52.27.155.124/pci/319379/42 : SUCCESS","accounts_in_message":[],"_revision_number":42},{"id":"b32425ffae760f0659624f49e8ecd2463ac6ea04","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-06 22:06:30.000000000","message":"Patch Set 42:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/42/check/check-dsvm-tempest-vz7-exe-minimal/7b145c9 : SUCCESS in 59m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":42},{"id":"aeb25311a0470cbe58fb61b37fb7d1648b8dc722","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-06 22:26:40.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/42/check/dsvm-tempest-neutron-network/34de5b7 : FAILURE in 2h 17m 20s","accounts_in_message":[],"_revision_number":42},{"id":"ece47b634814cb6ccde7f4d9f04aaf1760a6fcbd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-07 00:24:49.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.","accounts_in_message":[],"_revision_number":43},{"id":"56debb4733745be166da4d74bea534f76d65e437","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-07 01:34:15.000000000","message":"Patch Set 43:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/43/check/tempest-dsvm-full-xenial/cac18e6/ : SUCCESS in 59m 58s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"363c362cd32d04a6d1d65383aa031c65cdb33cb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 02:19:38.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/43/check/gate-nova-docs-ubuntu-xenial/9a72265//doc/build/html/ : SUCCESS in 3m 02s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-nova-pep8-ubuntu-xenial/4db7615/ : SUCCESS in 4m 20s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-nova-python27-db-ubuntu-xenial/0f00fec/ : SUCCESS in 8m 11s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/43/check/gate-nova-python34-db/fed9a4d/ : SUCCESS in 6m 49s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/43/check/gate-nova-python35-db-nv/bebeeb1/ : SUCCESS in 10m 10s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ad1552e/ : FAILURE in 50m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-grenade-dsvm-neutron-ubuntu-xenial/5e9edb1/ : SUCCESS in 54m 37s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-nova-tox-db-functional-ubuntu-xenial/e858c45/ : FAILURE in 45m 17s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f6ce77e/ : SUCCESS in 39m 41s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/43/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/526fd25/ : FAILURE in 41m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-cells-ubuntu-xenial/8e8671f/ : FAILURE in 1h 42m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/2574d21/ : SUCCESS in 1h 15m 56s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/e67f4c3/ : FAILURE in 45m 40s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/e37b29b/ : FAILURE in 1h 12m 30s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0b1bb1e/ : FAILURE in 21m 33s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/43/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/faf8c75/ : SUCCESS in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"5b9f791664c87332957fb80bbd0bc0356d696786","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 02:28:31.000000000","message":"Patch Set 43:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/43/check/check-dsvm-tempest-vz7-exe-minimal/33dccb4 : FAILURE in 55m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":43},{"id":"92fcaeb7521f643658e42369b15cad8e4f0cc169","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-07 03:05:40.000000000","message":"Patch Set 43:\n\n* pci-test http://52.27.155.124/pci/319379/43 : SUCCESS","accounts_in_message":[],"_revision_number":43},{"id":"7d26365f624bd987f18c3f618d85de4a55ad8187","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-07 03:45:19.000000000","message":"Patch Set 43:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/43/check/dsvm-tempest-xen/453c83e : FAILURE in 2h 28m 25s","accounts_in_message":[],"_revision_number":43},{"id":"021711cf258a1b8c5c07304a8b9eaf6d32445727","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-07 04:19:16.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":43},{"id":"adaccffc0eb6f6928b12742a3e2942daf272cb5b","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 04:19:25.000000000","message":"Patch Set 43:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":43},{"id":"9843ec6d2511026bb54afbb2f546dc71f2963900","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-07 04:20:57.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":43},{"id":"e9b69209492fcd524ee6ef10a74187fae44ad4ed","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 11:56:14.000000000","message":"Patch Set 43:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/43/ : SUCCESS in 1h 42m 11s","accounts_in_message":[],"_revision_number":43},{"id":"8c2ce417c71353caad7866d78ca5291ae00193a4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-07 14:45:37.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"d7970b09431e6198f37c06f64e800021129627d5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-07 14:47:46.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"ecab7ac574bf306e2b85a1588111fd1b5da19bcc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-07 16:21:17.000000000","message":"Patch Set 45:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/45/check/tempest-dsvm-full-xenial/91af36e/ : FAILURE in 1h 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"ec993e189430714dbc6ff69ebce1f991bc3a23b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 16:51:17.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/45/check/gate-nova-docs-ubuntu-xenial/8c3710c//doc/build/html/ : SUCCESS in 5m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-nova-pep8-ubuntu-xenial/7e5793e/ : SUCCESS in 5m 59s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-nova-python27-db-ubuntu-xenial/ee2d3c7/ : SUCCESS in 13m 05s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/45/check/gate-nova-python34-db/45a0d97/ : SUCCESS in 6m 52s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/45/check/gate-nova-python35-db-nv/d8c8a9a/ : SUCCESS in 16m 48s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/558cf7a/ : FAILURE in 1h 27m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ac2052c/ : SUCCESS in 35m 06s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-nova-tox-db-functional-ubuntu-xenial/6634053/ : FAILURE in 45m 29s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/7033508/ : SUCCESS in 1h 05m 53s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/45/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/697bcc0/ : FAILURE in 1h 07m 57s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-cells-ubuntu-xenial/b6f17cb/ : FAILURE in 1h 43m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d39a8a1/ : FAILURE in 56m 55s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/a7f7b92/ : FAILURE in 48m 31s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/a6c0fd0/ : FAILURE in 1h 03m 52s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b3cbe20/ : FAILURE in 35m 18s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/45/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/292e2ae/ : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"ee5f18b7b35b1f38055e619ccc1cf484b057a172","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-07 17:30:02.000000000","message":"Patch Set 44:\n\n* pci-test http://52.27.155.124/pci/319379/44 : SUCCESS","accounts_in_message":[],"_revision_number":44},{"id":"a137046430c85d684ad0a7f901a795f1f1a6436e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-07 17:49:19.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/45/check/dsvm-tempest-neutron-network/78ecc64 : SUCCESS in 2h 07m 21s","accounts_in_message":[],"_revision_number":45},{"id":"3341aea31817da0b2291ea794ee73fca3069fc9b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 18:03:58.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/45/check/check-dsvm-tempest-vz7-exe-minimal/baed72e : SUCCESS in 59m 39s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":45},{"id":"387d9c0e2af29e1c895e6f430806f5b4e8d1edbd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-07 18:17:47.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.","accounts_in_message":[],"_revision_number":46},{"id":"91a28deee2d05d4f7619bdc298fe257f37df3d0c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-07 18:20:44.000000000","message":"Patch Set 45:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":45},{"id":"5ebd9f1b53077a9c45d68b17d18766d1f4e278d0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-07 18:47:08.000000000","message":"Patch Set 45:\n\n* pci-test http://52.27.155.124/pci/319379/45 : SUCCESS","accounts_in_message":[],"_revision_number":45},{"id":"a125f7bad49b82e473974eaec56a1754ebf5f618","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-07 19:32:15.000000000","message":"Patch Set 46:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial tempest-dsvm-full-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"70ee9aafd12dffa938ef2de4ca1a11b71004e10c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-07 19:32:17.000000000","message":"Patch Set 46: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":46},{"id":"92014ad6afe1ae5db0aeb1a6c90d109e25a9479c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-07 19:32:20.000000000","message":"Patch Set 46:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":46},{"id":"e2665b08b3d0adc4c03b273b11d38f1f7326a566","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 19:32:24.000000000","message":"Patch Set 46:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":46},{"id":"50b77483fd0f9918ec0684a129fa1ac7256bb97f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-07 19:32:38.000000000","message":"Patch Set 46:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":46},{"id":"bf84ba02f2919ee9a6e06a5f715d98f252698ad6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-07 21:54:11.000000000","message":"Patch Set 46:\n\n* pci-test http://52.27.155.124/pci/319379/46 : SUCCESS","accounts_in_message":[],"_revision_number":46},{"id":"15695bd8ad1f0a03ad2134d0bb58fbf3c8550f60","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-07 22:15:41.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/46/check/dsvm-tempest-neutron-network/43e9f23 : FAILURE in 2h 07m 31s","accounts_in_message":[],"_revision_number":46},{"id":"eefd24888d351afe3b1c3f67f1802cb8852c4167","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-08 07:22:17.000000000","message":"Patch Set 46:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/46/check/dsvm-tempest-kvm/5d5f78f : SUCCESS in 1h 10m 12s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/46/check/dsvm-tempest-exe/a512088 : FAILURE in 43m 01s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":46},{"id":"d838a4338baca90af6ca96ba11eb731d29a49d61","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2016-12-08 10:58:24.000000000","message":"Patch Set 46:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/46/check/nova-pvm-dsvm-tempest-full/0d26a4e : FAILURE in 4h 01m 54s","accounts_in_message":[],"_revision_number":46},{"id":"0e4ce4f0312e495bb16ea6c32a98f46421818c36","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-09 18:25:06.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.","accounts_in_message":[],"_revision_number":47},{"id":"311c620644c305399885dc991f4dff18bdce81c7","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2016-12-09 19:34:58.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/47/ : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":47},{"id":"6de58269d9d8c3066549b13646cedb4ef28f6289","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-09 19:45:01.000000000","message":"Patch Set 47:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/47/check/tempest-dsvm-full-xenial/9b2f225/ : FAILURE in 1h 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"b746505ab5993b52b4017d46535d175dfb216f31","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-09 20:42:33.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/47/check/check-dsvm-tempest-vz7-exe-minimal/d46fa9e : SUCCESS in 1h 05m 04s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":47},{"id":"87ddd248bfc8a2ffd455a7ea54274ae49d7850e1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-09 20:42:43.000000000","message":"Patch Set 47:\n\n* pci-test http://52.27.155.124/pci/319379/47 : SUCCESS","accounts_in_message":[],"_revision_number":47},{"id":"670d30640905dc4c71fe773ffd8aac0a7cdf00c9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-09 21:13:47.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/47/check/dsvm-tempest-neutron-network/c275e5d : FAILURE in 2h 03m 58s","accounts_in_message":[],"_revision_number":47},{"id":"ee1a048697e98affeb8dcdc3d4ded487fb5eb579","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-09 21:21:09.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/47/check/dsvm-tempest-xen/658d0ca : SUCCESS in 2h 23m 34s","accounts_in_message":[],"_revision_number":47},{"id":"d702233c32112bb8e8fcb7a1d40ad164431d0641","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-09 23:09:29.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/47/check/dsvm-tempest-kvm/fdcd00e : SUCCESS in 1h 05m 50s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/47/check/dsvm-tempest-exe/7a477fd : FAILURE in 42m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":47},{"id":"f75439d97a06767be8a885d0f9cba950d79b2778","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 23:24:38.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/47/check/gate-nova-docs-ubuntu-xenial/6c4693d//doc/build/html/ : SUCCESS in 4m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-nova-pep8-ubuntu-xenial/33946cd/ : SUCCESS in 3m 47s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-nova-python27-db-ubuntu-xenial/2994a11/ : SUCCESS in 7m 37s\n- gate-nova-python34-db http://logs.openstack.org/79/319379/47/check/gate-nova-python34-db/7c95a9d/ : SUCCESS in 10m 47s\n- gate-nova-python35-db-nv http://logs.openstack.org/79/319379/47/check/gate-nova-python35-db-nv/7a3d826/ : SUCCESS in 9m 50s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/83a710e/ : SUCCESS in 52m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-grenade-dsvm-neutron-ubuntu-xenial/32f65ac/ : SUCCESS in 34m 00s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-nova-tox-db-functional-ubuntu-xenial/c1bf5f9/ : FAILURE in 45m 25s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/dcb95b5/ : SUCCESS in 41m 22s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/47/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2a2960f/ : FAILURE in 1h 11m 58s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-cells-ubuntu-xenial/f567704/ : FAILURE in 1h 30m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/83477ce/ : SUCCESS in 1h 04m 08s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/53dda2a/ : FAILURE in 52m 08s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/b15a2ef/ : SUCCESS in 2h 22m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/50ccf29/ : FAILURE in 34m 11s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/47/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/a3f0a6b/ : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"1a102bfebafca469440a1e6eeb4572b9380a90dd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-12-09 23:27:13.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/47/check/tempest-dsvm-full-nfv-xenial/3fef3ca : SUCCESS in 2h 22m 12s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/47/check/tempest-dsvm-intel-nfv-xenial/6b9e0e1 : FAILURE in 44m 25s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/47/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/adfa89f : FAILURE in 1h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"c5c9ff198ab652dc001690ac7ed74f80888393fd","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2016-12-10 10:20:02.000000000","message":"Patch Set 47:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/47/check/nova-pvm-dsvm-tempest-full/c9f1653 : FAILURE in 3h 18m 51s","accounts_in_message":[],"_revision_number":47},{"id":"1e94fbfd84b1dd9bf1a08ca0459d2f735d4022f8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-14 00:18:07.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"a0dc674c7150a45acf6306c760131e29b5c747fa","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-14 00:34:48.000000000","message":"Patch Set 48:\n\n* pci-test http://52.27.155.124/pci/319379/48 : FAILURE","accounts_in_message":[],"_revision_number":48},{"id":"1d41daec3b037d92d4a2528592b57b51b940ec4c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-14 01:32:25.000000000","message":"Patch Set 48:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/48/check/tempest-dsvm-full-xenial/30f340a/ : FAILURE in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"8e93ab3a81267accac4803c75a9527f9e62be112","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2016-12-14 01:35:23.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/48/ : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":48},{"id":"8547f939d90cc78c48b01e8c77ac04f7a3a4661a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-14 01:48:38.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/48/check/gate-nova-docs-ubuntu-xenial/b9a20df//doc/build/html/ : SUCCESS in 4m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-nova-pep8-ubuntu-xenial/0deb806/ : SUCCESS in 3m 31s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-nova-python27-db-ubuntu-xenial/1bc3052/ : SUCCESS in 10m 23s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/48/check/gate-nova-python35-db/c024d56/ : SUCCESS in 7m 19s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/42344e0/ : FAILURE in 52m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4a43d60/ : SUCCESS in 34m 43s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-nova-tox-db-functional-ubuntu-xenial/9f160ae/ : FAILURE in 10m 49s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/cad9a71/ : SUCCESS in 36m 42s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/48/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/48f687d/ : FAILURE in 41m 52s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-cells-ubuntu-xenial/04bb770/ : FAILURE in 1h 28m 01s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ba30ba7/ : FAILURE in 1h 15m 06s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/35b2422/ : FAILURE in 48m 34s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f36bd00/ : FAILURE in 50m 17s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a39048c/ : FAILURE in 23m 22s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/48/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/00cd4ee/ : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"d1a833a92d6a8224650a9d734d89eb20f47cd95a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-12-14 01:54:36.000000000","message":"Patch Set 48:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1645530\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":48},{"id":"a8d5bd158b96e292507b2a1c6ef978b7e819cf09","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-14 03:01:35.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/48/check/dsvm-tempest-xen/00a7fc6 : SUCCESS in 2h 40m 46s","accounts_in_message":[],"_revision_number":48},{"id":"1648ee0f1eeadc88ad5f406bdb5ba956a5171119","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-14 03:08:57.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/48/check/check-dsvm-tempest-vz7-exe-minimal/dfd8009 : SUCCESS in 1h 00m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":48},{"id":"a11d6dab333ae736662c37bee044f104dbb2e56f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-14 06:09:23.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/48/check/dsvm-tempest-kvm/f63d4c2 : SUCCESS in 1h 06m 45s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/48/check/dsvm-tempest-exe/eb7cfcf : FAILURE in 42m 16s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":48},{"id":"d00a8d23261c90f10f1c14526e67c16b18db9418","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-12-14 06:53:51.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/48/check/tempest-dsvm-full-nfv-xenial/bab6c7f : SUCCESS in 1h 55m 49s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/48/check/tempest-dsvm-intel-nfv-xenial/6d77103 : SUCCESS in 44m 11s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/48/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/6a69fb7 : FAILURE in 1h 36m 24s","accounts_in_message":[],"_revision_number":48},{"id":"55302b82362ae71a34564b6596645d8bf4f3f793","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-14 09:08:22.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/48/check-nova/Nova-ML2-Sriov/7be031d : SUCCESS in 48m 58s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/48/check-nova/Nova-MACVTAP-ML2-Sriov/3ef0188 : SUCCESS in 44m 02s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":48},{"id":"1c381e6830144f923c959cf77b9f80ba28a914ef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-14 10:53:59.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/48/check/dsvm-tempest-neutron-network/dd1b1d7 : SUCCESS in 2h 14m 55s","accounts_in_message":[],"_revision_number":48},{"id":"e3b1ecc86906e3d9d17eb27dcf387adf560ef85a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-14 11:40:30.000000000","message":"Patch Set 48:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/48/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/d92fd97 : SUCCESS in 59m 20s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":48},{"id":"9e8034e67f22666b2d1a7fe483cb1cac2eb3a9c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-12-14 19:50:27.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"417fead1f9a8da396211661fad763b130f69a211","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-12-14 20:56:16.000000000","message":"Patch Set 49:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/49/check/tempest-dsvm-full-xenial/edd9d44/ : FAILURE in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"2eba6e4dbf14e85c34f2a5b3f88044ef45ca5326","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2016-12-14 21:15:56.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/49/ : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":49},{"id":"da0a5f7255ab952e957a62bdb3e94e052f8d9b83","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-14 21:26:43.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/49/check/gate-nova-docs-ubuntu-xenial/0f255de//doc/build/html/ : SUCCESS in 4m 35s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-nova-pep8-ubuntu-xenial/bd93c94/ : SUCCESS in 4m 22s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-nova-python27-db-ubuntu-xenial/7b32c1e/ : SUCCESS in 14m 35s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/49/check/gate-nova-python35-db/67927c4/ : SUCCESS in 7m 50s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8b4bb2d/ : FAILURE in 52m 27s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d17b0e0/ : SUCCESS in 55m 07s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-nova-tox-db-functional-ubuntu-xenial/2ff48af/ : SUCCESS in 11m 32s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a9ff2ad/ : SUCCESS in 38m 38s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/49/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/4d4c2e8/ : FAILURE in 1h 06m 03s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-cells-ubuntu-xenial/11e1df2/ : FAILURE in 1h 26m 58s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0b63f67/ : FAILURE in 1h 19m 28s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/1ae0669/ : FAILURE in 50m 51s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/97514be/ : FAILURE in 53m 44s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/ecedea9/ : FAILURE in 23m 12s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/49/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8364bfe/ : SUCCESS in 51m 47s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"0b25a72c826b09beeb61b09e9afa0c83a78f9e4c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-12-14 21:29:53.000000000","message":"Patch Set 49:\n\n* pci-test http://52.27.155.124/pci/319379/49 : SUCCESS","accounts_in_message":[],"_revision_number":49},{"id":"d3a3753fae4f9e334f4e123792f26577199bd4a6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-12-14 23:40:04.000000000","message":"Patch Set 49:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/49/check/check-dsvm-tempest-vz7-exe-minimal/3a2d5dc : FAILURE in 12m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":49},{"id":"3787cd465cb696686690926bfa36188a88816cd9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-12-14 23:59:01.000000000","message":"Patch Set 49:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/49/check/dsvm-tempest-xen/53ab595 : FAILURE in 2h 21m 58s","accounts_in_message":[],"_revision_number":49},{"id":"2370a2fd1d0d80ba48fd2211a663cc83efac48cb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-12-15 00:38:11.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/49/check/tempest-dsvm-full-nfv-xenial/35c3942 : FAILURE in 1h 57m 43s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/49/check/tempest-dsvm-intel-nfv-xenial/baa34b3 : SUCCESS in 45m 50s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/49/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/79c2228 : FAILURE in 1h 39m 46s","accounts_in_message":[],"_revision_number":49},{"id":"9275ea256b757f30fc64cee4e3212cbb55ebea85","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-15 01:41:14.000000000","message":"Patch Set 49:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/49/check/dsvm-tempest-neutron-network/28b5574 : FAILURE in 18m 34s","accounts_in_message":[],"_revision_number":49},{"id":"629b90da9b4cccf82edbc22e5ca25e4287ad0a1f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-12-15 05:30:21.000000000","message":"Patch Set 49:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/49/check/dsvm-tempest-kvm/699ccf5 : FAILURE in 10m 57s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/49/check/dsvm-tempest-exe/4e9c4c9 : FAILURE in 32s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":49},{"id":"23acb6e50934e2282b7269b11c18c329268ff869","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-12-15 06:51:53.000000000","message":"Patch Set 49:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/49/check/dsvm-tempest-neutron-network/714839e : FAILURE in 2h 17m 11s","accounts_in_message":[],"_revision_number":49},{"id":"4931d29c19bf22cd4bd36de3ceec422752d65bab","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-15 21:14:35.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/49/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/cb01032 : SUCCESS in 52m 12s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":49},{"id":"0c664e3c049e930b8163815d0f75ed8c54d43b36","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-12-15 21:15:06.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/49/check-nova/Nova-ML2-Sriov/e67ef52 : SUCCESS in 37m 30s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/49/check-nova/Nova-MACVTAP-ML2-Sriov/297671b : SUCCESS in 44m 51s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":49},{"id":"a8b9768c5f4f5247dc11b6c50e201665984efb96","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-03 19:41:02.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.","accounts_in_message":[],"_revision_number":50},{"id":"24e867c9551e3685cce2ab875f91e5e4425b7977","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-03 20:40:15.000000000","message":"Patch Set 50:\n\n* pci-test http://52.27.155.124/pci/319379/50 : SUCCESS","accounts_in_message":[],"_revision_number":50},{"id":"87a010cf4177203b1e614197fb2a4566997f28bf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-03 20:46:10.000000000","message":"Patch Set 50: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/50/check/tempest-dsvm-full-xenial/18430cb/ : FAILURE in 1h 02m 01s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/50/check/grenade-dsvm-xenial/481cc4d/ : SUCCESS in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"af30662b374793f826acab5ebedd25fce2748f64","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-03 20:54:11.000000000","message":"Patch Set 50:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/50/ : FAILURE in 1h 09m 34s","accounts_in_message":[],"_revision_number":50},{"id":"1ea1299443adeba62870537a4b1a074c803cca54","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-03 21:35:31.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.","accounts_in_message":[],"_revision_number":51},{"id":"74d724137b77156c7620feaaca50fded6f2572a8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-03 22:43:45.000000000","message":"Patch Set 51: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/51/check/tempest-dsvm-full-xenial/9421bc0/ : FAILURE in 1h 05m 20s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/51/check/grenade-dsvm-xenial/d2b0b6e/ : SUCCESS in 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"9d910ab8ce96e79a9c9e43cb90e33e1e054616e9","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-03 23:01:56.000000000","message":"Patch Set 51:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/51/ : FAILURE in 1h 12m 42s","accounts_in_message":[],"_revision_number":51},{"id":"eb3ce96bfc650d8ed7f99cd1aaafc8fc5d538abb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-03 23:35:27.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/51/check/gate-nova-docs-ubuntu-xenial/1cf155d//doc/build/html/ : SUCCESS in 2m 46s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-nova-pep8-ubuntu-xenial/f9a7203/ : SUCCESS in 3m 58s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-nova-python27-db-ubuntu-xenial/544cb2e/ : FAILURE in 7m 38s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/51/check/gate-nova-python35-db/ee9e94b/ : FAILURE in 7m 51s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a2126f6/ : SUCCESS in 58m 02s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8e7dc66/ : SUCCESS in 36m 51s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-nova-tox-db-functional-ubuntu-xenial/a54a69a/ : SUCCESS in 10m 42s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/7707965/ : SUCCESS in 38m 57s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/51/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/7ad6deb/ : FAILURE in 39m 16s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-cells-ubuntu-xenial/e3df90b/ : FAILURE in 1h 55m 26s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a91174a/ : FAILURE in 57m 52s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/c7d01ba/ : FAILURE in 51m 51s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/9a2e1bb/ : FAILURE in 53m 47s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/e0f14bb/ : FAILURE in 24m 41s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/51/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4dae6d5/ : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"9018c28f03b685bc64a3a9045b11855c0d40ab86","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-03 23:37:47.000000000","message":"Patch Set 51:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python35-db: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1509004\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":51},{"id":"7d5d93ad48a0cc4064a52871340be206ec58c98b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-03 23:49:16.000000000","message":"Patch Set 51:\n\n* pci-test http://52.27.155.124/pci/319379/51 : SUCCESS","accounts_in_message":[],"_revision_number":51},{"id":"dbd7dc1d0d40ff0dcc8e747667fc4e2da6c382c2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 00:08:20.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/51/check/dsvm-tempest-kvm/27c388a : FAILURE in 57m 34s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/51/check/dsvm-tempest-exe/7d999f9 : SUCCESS in 45m 32s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":51},{"id":"0445f7566663b733edd35cd7bc56ea7fc1535cf4","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-04 01:11:15.000000000","message":"Patch Set 51:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/51/check/nova-pvm-dsvm-tempest-full/5e3b198 : FAILURE in 3h 34m 28s","accounts_in_message":[],"_revision_number":51},{"id":"788c48f4f7a9ee795dec315c11205d99485289fb","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-01-04 01:22:25.000000000","message":"Patch Set 51:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/51/check/dsvm-tempest-xen/ce31922 : FAILURE in 2h 14m 47s","accounts_in_message":[],"_revision_number":51},{"id":"31b26bf5bdf0b20ba6e9f18978e6157cfec20e3e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 02:08:11.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/51/check/tempest-dsvm-full-nfv-xenial/a63d134 : FAILURE in 2h 00m 16s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/51/check/tempest-dsvm-intel-nfv-xenial/da8b96d : SUCCESS in 41m 41s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/51/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/d1c4a0e : FAILURE in 1h 28m 55s","accounts_in_message":[],"_revision_number":51},{"id":"0f735e6126a39ea5fe30cca12e94b11808ddc2b6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-04 04:46:15.000000000","message":"Patch Set 51:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/51/check/dsvm-tempest-neutron-network/21cc6c4 : FAILURE in 1h 58m 30s","accounts_in_message":[],"_revision_number":51},{"id":"6fae9e7dd9004e0e0d614f140a653557b9d0e71b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 04:58:59.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/51/check/check-dsvm-tempest-vz7-exe-minimal/018a477 : SUCCESS in 57m 34s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/51/check/check-dsvm-tempest-vz7-exe-multinode/e7f20fb : FAILURE in 50m 27s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":51},{"id":"a55ff55ce184d4cf231d852323f2126b2a718a1c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-04 09:02:37.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/51/check-nova/Nova-ML2-Sriov/510d6aa : SUCCESS in 42m 20s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/51/check-nova/Nova-MACVTAP-ML2-Sriov/67e8312 : SUCCESS in 45m 37s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":51},{"id":"45a37b7e9ffc9fd6ddcbe298a4342af3f1001a76","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-04 09:26:32.000000000","message":"Patch Set 51:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/51/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/79e3eb6 : SUCCESS in 1h 00m 20s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":51},{"id":"d018f2a7fe4a2e95ab2fc873c41f8d1f2e7f0135","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-04 18:10:47.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"f87ccdfdebd9fe3512a4eef41dc10fd28e6be1a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-04 19:13:14.000000000","message":"Patch Set 52: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/52/check/tempest-dsvm-full-xenial/7c416ca/ : FAILURE in 59m 58s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/52/check/grenade-dsvm-xenial/d141f35/ : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"f0806b112efd88cc2f80365e82f976cc2baec93a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-04 19:13:40.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/52/check/gate-nova-docs-ubuntu-xenial/ecdf5a3//doc/build/html/ : SUCCESS in 3m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-nova-pep8-ubuntu-xenial/0603e09/ : SUCCESS in 3m 44s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-nova-python27-db-ubuntu-xenial/437aa98/ : FAILURE in 10m 45s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/52/check/gate-nova-python35-db/bbe3562/ : FAILURE in 8m 03s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/74fde35/ : SUCCESS in 57m 04s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4c8156f/ : SUCCESS in 32m 51s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-nova-tox-db-functional-ubuntu-xenial/cbf56a9/ : SUCCESS in 11m 03s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/4dbb907/ : SUCCESS in 37m 52s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/52/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/aca17bd/ : FAILURE in 41m 33s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-cells-ubuntu-xenial/e066224/ : FAILURE in 26m 37s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/912a4c0/ : SUCCESS in 54m 19s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3e10540/ : FAILURE in 55m 42s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/5cc9bb9/ : FAILURE in 58m 11s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/9ad43af/ : FAILURE in 43m 14s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/52/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4e37c83/ : SUCCESS in 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"aa0ff7321c67c967314e6f0f2eb2478edb4c5672","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-04 19:50:17.000000000","message":"Patch Set 52:\n\n* pci-test http://52.27.155.124/pci/319379/52 : SUCCESS","accounts_in_message":[],"_revision_number":52},{"id":"6ceb8a4eb893e201a1465ad7a56bb60db1f0aaa9","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-04 20:02:30.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/52/ : SUCCESS in 1h 49m 02s","accounts_in_message":[],"_revision_number":52},{"id":"71388985f41cca301a2b973c1954ed42bada2163","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-04 21:48:13.000000000","message":"Patch Set 52:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/52/check/nova-pvm-dsvm-tempest-full/15cb93f : FAILURE in 3h 34m 42s","accounts_in_message":[],"_revision_number":52},{"id":"fe94d31169f3ba32bbd941f0439122144bd13dce","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-01-04 22:13:35.000000000","message":"Patch Set 52:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":52},{"id":"78500654a4003d22c7f102e1077325ea86c1ecd1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 22:14:03.000000000","message":"Patch Set 52:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":52},{"id":"229873a52f92ec2a322ac8b106d1417a5c270dc2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-04 22:14:04.000000000","message":"Patch Set 52:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":52},{"id":"e9058e63b2476c9dd868dda2261d609de80ab8ca","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 22:14:04.000000000","message":"Patch Set 52: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":52},{"id":"daf8026d3934b7fd65ad8412c8faa32b3a1ee4fc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-04 22:14:13.000000000","message":"Patch Set 52:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":52},{"id":"75c0106e251b7f70335d1b1268370b5cc431a7e8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-04 22:15:24.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":52},{"id":"1c8a32c70817e316528c2b12357e0181c233eb39","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-04 22:16:55.000000000","message":"Patch Set 52:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":52},{"id":"77d8afdf6a8c1632503686c6b671c21b0b2b468b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-04 23:30:03.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.","accounts_in_message":[],"_revision_number":53},{"id":"55f4f87d54f5f01cb096170db9576a375dfb2b7a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-05 00:29:37.000000000","message":"Patch Set 53: Verified+1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/53/check/tempest-dsvm-full-xenial/d37bf06/ : SUCCESS in 56m 54s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/53/check/grenade-dsvm-xenial/dd4d0c7/ : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"018e2e666177d04aa17a42c14c1f360de7af7add","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-05 00:33:50.000000000","message":"Patch Set 53:\n\n* pci-test http://52.27.155.124/pci/319379/53 : SUCCESS","accounts_in_message":[],"_revision_number":53},{"id":"dc6fce00a4ed9796bcc7e708dd5509576e47336f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-05 00:49:51.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/53/check/gate-nova-docs-ubuntu-xenial/555130c//doc/build/html/ : SUCCESS in 4m 06s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-nova-pep8-ubuntu-xenial/2b0fb2d/ : SUCCESS in 6m 37s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-nova-python27-db-ubuntu-xenial/0611fdb/ : FAILURE in 10m 26s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/53/check/gate-nova-python35-db/d6fd9d4/ : FAILURE in 6m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/99d627f/ : FAILURE in 1h 16m 05s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-grenade-dsvm-neutron-ubuntu-xenial/02dc741/ : SUCCESS in 32m 51s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-nova-tox-db-functional-ubuntu-xenial/37ceedf/ : SUCCESS in 10m 21s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b94af68/ : SUCCESS in 37m 52s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/53/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/849f9e0/ : FAILURE in 39m 11s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-cells-ubuntu-xenial/a5217cc/ : FAILURE in 15m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5673845/ : SUCCESS in 58m 48s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/0fe71fd/ : FAILURE in 52m 04s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/7f79a35/ : FAILURE in 53m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0f6170d/ : FAILURE in 23m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/53/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/679c010/ : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"4dc8e9f46727ccffe09075c44cf1a93f4d3a6b4e","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-05 00:52:01.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/53/ : SUCCESS in 1h 19m 05s","accounts_in_message":[],"_revision_number":53},{"id":"3eba52d946cb57d504f699e62880e2dc342c8bb6","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-05 03:10:38.000000000","message":"Patch Set 53:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/53/check/nova-pvm-dsvm-tempest-full/632e332 : FAILURE in 3h 35m 38s","accounts_in_message":[],"_revision_number":53},{"id":"f415587a1248d30cd591f0fe3e5b79446454f149","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-05 04:44:37.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/53/check/dsvm-tempest-kvm/0d14291 : SUCCESS in 1h 03m 07s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/53/check/dsvm-tempest-exe/0be0aa1 : SUCCESS in 43m 17s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":53},{"id":"e0413009b9018da8a1946b3b2e20abfec7fa7897","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-05 06:26:15.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-full-nfv-xenial/10ebd2d : UNSTABLE in 1h 47m 12s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-intel-nfv-xenial/c5bb203 : UNSTABLE in 39m 05s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/fdc01c4 : FAILURE in 1h 40m 23s","accounts_in_message":[],"_revision_number":53},{"id":"e47bcfad79707c0b5b9421917cf97959e42c1d50","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-05 07:24:35.000000000","message":"Patch Set 53:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/53/check/dsvm-tempest-neutron-network/124b8a3 : FAILURE in 3h 54m 40s","accounts_in_message":[],"_revision_number":53},{"id":"236aef34a4552969dd15b5c8e46653fd34b849cd","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-01-05 07:34:42.000000000","message":"Patch Set 53:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/79/319379/53/check/dsvm-tempest-xen/71398f5 : FAILURE in 2h 33m 29s","accounts_in_message":[],"_revision_number":53},{"id":"2b1329708f353fc949132299b7dbc1dc1dc086c5","author":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"date":"2017-01-05 07:54:26.000000000","message":"Patch Set 53:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":53},{"id":"827039140a97d2d95cae897d94ef2e04418056b8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-05 10:19:00.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-full-nfv-xenial/95716db : FAILURE in 1h 51m 28s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-intel-nfv-xenial/607ee96 : SUCCESS in 43m 47s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/79/319379/53/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/7e250f4 : FAILURE in 1h 31m 31s","accounts_in_message":[],"_revision_number":53},{"id":"fd16bf2b4ec1b96cac4b15aefcaf5823e65eca5f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-05 11:32:43.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/53/check-nova/Nova-ML2-Sriov/5b35104 : SUCCESS in 45m 35s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/53/check-nova/Nova-MACVTAP-ML2-Sriov/46bca74 : SUCCESS in 45m 46s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":53},{"id":"6df8e44d745439a8f008cbc0661df3aca19602de","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-05 15:14:57.000000000","message":"Patch Set 53:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":53},{"id":"8c2fdcea37e28ffec88609329264368091bd3948","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-05 15:15:11.000000000","message":"Patch Set 53:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":53},{"id":"b8feb752b30e466813830f408800ba1baeeb2966","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-07 21:06:23.000000000","message":"Patch Set 50:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/50/1460/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":50},{"id":"127315a37b44a6c95ced659a84aa2ad8718e726e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-08 02:26:09.000000000","message":"Patch Set 51:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/51/1488/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":51},{"id":"ac0b6159f5f346d43adc12d433e2ddcf85db4a57","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-09 07:23:49.000000000","message":"Patch Set 52:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/52/1637/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":52},{"id":"214e8cf9884fd2f6a41637c41e695098d6e206c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-09 14:18:33.000000000","message":"Patch Set 53:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/53/1666/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":53},{"id":"fdfa05bbd7b13f24f39439b3fa98e1425bf79c9a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-12 15:37:48.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"04fbf2f48805aded8662204da1d3803c2e824ee3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-12 16:11:37.000000000","message":"Patch Set 54:\n\n* pci-test http://52.27.155.124/pci/319379/54 : SUCCESS","accounts_in_message":[],"_revision_number":54},{"id":"49cb1c474e0376bb12c5510372c42d90f18a4ad4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-12 16:17:56.000000000","message":"Patch Set 54: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":54},{"id":"a02f7ab4db985073001bbbd8b6e2ee102a99c596","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:00.000000000","message":"Patch Set 54:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":54},{"id":"a2395c436bccc7a8a71c6f57f3b94cfa6aebe858","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-12 16:18:00.000000000","message":"Patch Set 54:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":54},{"id":"b801e2ad56e5b39af48da6bcfec7651465488748","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:01.000000000","message":"Patch Set 54:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":54},{"id":"0a318cf5eaad0b8a1171d4c1c2b13aa4aaaf099a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:06.000000000","message":"Patch Set 54:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":54},{"id":"8450b84ba4cda78ff171ad674333b9eedf00f5c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 16:18:09.000000000","message":"Patch Set 54: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":54},{"id":"fd2e4f4c7978ec8a811ec8c22835132ab576f850","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:10.000000000","message":"Patch Set 54:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":54},{"id":"a37341609e6606219e4b08cdc6a8a3169a617d07","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:30.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":54},{"id":"3afa9e8413b48b5fcceacf08a6caac7218476c39","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-12 16:18:51.000000000","message":"Patch Set 54:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":54},{"id":"161848c745176939626b72a6dece58ecef0c2beb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-12 16:19:48.000000000","message":"Patch Set 54:\n\nrecheck had to bump a dep","accounts_in_message":[],"_revision_number":54},{"id":"57f7f7d0459f29f2f1daef1161ea6f657e48a6cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-12 17:20:43.000000000","message":"Patch Set 54: Verified+1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/54/check/tempest-dsvm-full-xenial/279be14/ : SUCCESS in 1h 00m 16s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/54/check/grenade-dsvm-xenial/8ed2b41/ : SUCCESS in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"b9baa80f72c0591176dab877d24da2feffc70b70","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 18:09:57.000000000","message":"Patch Set 54:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/54/check/gate-nova-docs-ubuntu-xenial/c3fccae//doc/build/html/ : SUCCESS in 5m 17s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-nova-pep8-ubuntu-xenial/4387851/ : SUCCESS in 6m 40s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-nova-python27-db-ubuntu-xenial/63b0084/ : FAILURE in 11m 24s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/54/check/gate-nova-python35-db/f8afca1/ : FAILURE in 11m 14s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4e6ffca/ : SUCCESS in 1h 46m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-grenade-dsvm-neutron-ubuntu-xenial/739641f/ : SUCCESS in 32m 33s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-nova-tox-db-functional-ubuntu-xenial/6ac1e50/ : FAILURE in 45m 29s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/92b36e8/ : FAILURE in 18m 25s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/d254e7c/ : SUCCESS in 37m 38s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/54/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/89724ed/ : SUCCESS in 36m 12s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-cells-ubuntu-xenial/4fe7343/ : FAILURE in 32m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/371017b/ : SUCCESS in 59m 11s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/a263135/ : FAILURE in 1h 25m 45s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/5c7f1d1/ : SUCCESS in 1h 04m 44s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/20966fd/ : FAILURE in 36m 55s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/54/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/32532e8/ : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"4284f9f1f3d8fc1bccbdf26839f6c54685006978","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-12 18:30:52.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"39263f3f7c9dc4a39e5bacb1ea32c34a784d1683","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-12 19:09:39.000000000","message":"Patch Set 55:\n\n* pci-test http://52.27.155.124/pci/319379/55 : SUCCESS","accounts_in_message":[],"_revision_number":55},{"id":"102077d7473d7d2a6ff8b05545c0ccf348cab396","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-12 19:20:50.000000000","message":"Patch Set 55: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/55/check/tempest-dsvm-full-xenial/87c2921/ : FAILURE in 23m 11s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/55/check/grenade-dsvm-xenial/7f1b193/ : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"f468bdabd44a572886317937aa2b9138cc905fdb","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-12 19:41:19.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/55/check/check-dsvm-tempest-vz7-exe-minimal/3339890 : SUCCESS in 58m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":55},{"id":"a6549beb8d7206a72c486bc0c18876cd03570ce9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 19:57:11.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/55/check/gate-nova-docs-ubuntu-xenial/3915a52//doc/build/html/ : SUCCESS in 3m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-nova-pep8-ubuntu-xenial/9b52685/ : SUCCESS in 3m 49s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-nova-python27-db-ubuntu-xenial/c672371/ : FAILURE in 12m 39s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/55/check/gate-nova-python35-db/a456c21/ : FAILURE in 8m 38s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a6a953f/ : FAILURE in 1h 00m 10s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-grenade-dsvm-neutron-ubuntu-xenial/6439f25/ : SUCCESS in 35m 16s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-nova-tox-db-functional-ubuntu-xenial/acb4e9a/ : SUCCESS in 16m 16s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/e93d9c5/ : SUCCESS in 13m 33s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e4cb576/ : SUCCESS in 1h 05m 22s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/55/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/49b8a2c/ : SUCCESS in 56m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-cells-ubuntu-xenial/b8611f0/ : FAILURE in 33m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4d63759/ : SUCCESS in 59m 32s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/deae515/ : FAILURE in 1h 18m 59s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/ce08a7c/ : FAILURE in 55m 20s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/88976e5/ : FAILURE in 42m 23s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/55/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/d794836/ : FAILURE in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"5c5b0056a0e5da69b765322d18e540617c6bd579","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-12 20:41:51.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":55},{"id":"eff9c71af258ca6e6408edb9e5968d342fa7a88f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-12 20:47:30.000000000","message":"Patch Set 55:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/55/check/dsvm-tempest-neutron-network/bedcfa5 : FAILURE in 1h 46m 40s","accounts_in_message":[],"_revision_number":55},{"id":"8ae302da25452c60b47977a1707b1fda5efed4b4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-12 21:28:50.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/55/check-nova/Nova-ML2-Sriov/42891a6 : SUCCESS in 45m 54s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/55/check-nova/Nova-MACVTAP-ML2-Sriov/0ddd413 : SUCCESS in 42m 40s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":55},{"id":"289eb02abd1e7105fdda775574e50e218962bc8d","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-12 21:34:12.000000000","message":"Patch Set 55:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/55/ : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":55},{"id":"205db393d96d6e17efe32543c787e39f689d5ea3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-12 22:20:14.000000000","message":"Patch Set 55:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/55/check/nova-pvm-dsvm-tempest-full/6841134 : FAILURE in 3h 48m 22s","accounts_in_message":[],"_revision_number":55},{"id":"edb4752b4d132a7f43af6f05dfaa91c3091c9934","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-12 22:21:16.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.","accounts_in_message":[],"_revision_number":56},{"id":"e044bcf5a802794cf174c37f1867567083546900","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-12 22:33:05.000000000","message":"Patch Set 56:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/56/ : FAILURE in 10m 03s","accounts_in_message":[],"_revision_number":56},{"id":"846c06e89a36b5f5a790bf0be47ce58491f70093","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-12 23:14:52.000000000","message":"Patch Set 56:\n\n* pci-test http://52.27.155.124/pci/319379/56 : SUCCESS","accounts_in_message":[],"_revision_number":56},{"id":"7a1858dced41834b5c02f81ebaa63011182a97dc","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-12 23:29:55.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/56/check/check-dsvm-tempest-vz7-exe-minimal/de431e0 : SUCCESS in 58m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":56},{"id":"4ebd9045bc816473a9f7c9aa53dbf9b89fde61c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 23:55:49.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/56/check/gate-nova-docs-ubuntu-xenial/cbd93e5//doc/build/html/ : SUCCESS in 3m 15s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-nova-pep8-ubuntu-xenial/4160608/ : SUCCESS in 6m 38s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-nova-python27-db-ubuntu-xenial/e728149/ : SUCCESS in 7m 29s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/56/check/gate-nova-python35-db/e146a20/ : SUCCESS in 7m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/321094c/ : SUCCESS in 1h 02m 50s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-grenade-dsvm-neutron-ubuntu-xenial/569c64f/ : SUCCESS in 48m 24s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-nova-tox-db-functional-ubuntu-xenial/a3d01ab/ : SUCCESS in 11m 22s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/1a1df50/ : SUCCESS in 8m 06s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/221eab3/ : SUCCESS in 39m 19s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/56/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/d1a5bf6/ : SUCCESS in 37m 42s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-cells-ubuntu-xenial/8d0e7d2/ : FAILURE in 32m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/98475d9/ : SUCCESS in 1h 05m 08s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/52e7019/ : FAILURE in 1h 28m 40s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/eb46308/ : SUCCESS in 1h 09m 59s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/2133ee7/ : FAILURE in 28m 04s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/56/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f660187/ : SUCCESS in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"13540bfdbdde875010bde8b7689dc5dc5ecfa8d6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-13 00:06:41.000000000","message":"Patch Set 56:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/56/check/dsvm-tempest-neutron-network/ee13740 : FAILURE in 1h 39m 07s","accounts_in_message":[],"_revision_number":56},{"id":"4a847d329c889245630df66ec5b25d0caad0439a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-13 00:15:29.000000000","message":"Patch Set 56: Verified+1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/56/check/tempest-dsvm-full-xenial/c4b5061/ : SUCCESS in 1h 02m 50s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/56/check/grenade-dsvm-xenial/b0ed07c/ : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"5616d2738851d990698838d6c6cdab8d95792f64","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-13 02:09:13.000000000","message":"Patch Set 56:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/56/check/nova-pvm-dsvm-tempest-full/81fcb6f : FAILURE in 3h 46m 38s","accounts_in_message":[],"_revision_number":56},{"id":"219f5ff02efcc4aa23af7be3f0b5e5dbd74fde3e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-13 02:41:56.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/56/check-nova/Nova-ML2-Sriov/7021e43 : SUCCESS in 47m 03s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/56/check-nova/Nova-MACVTAP-ML2-Sriov/861dc41 : SUCCESS in 47m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":56},{"id":"cf8c687afb6efca93c4ebe76a688ab3f1292d003","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-13 03:36:38.000000000","message":"Patch Set 56:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/56/check/dsvm-tempest-kvm/79a52f0 : FAILURE in 1h 02m 06s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/56/check/dsvm-tempest-exe/69f324c : SUCCESS in 42m 38s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":56},{"id":"8391120d9b061810a2a363da5c9d30e291f88215","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-13 04:51:06.000000000","message":"Patch Set 56:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/56/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/cb06e6d : SUCCESS in 52m 15s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":56},{"id":"dffbba7e904ce26efa09f702c8d7d4f7bf09ec98","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-13 05:23:46.000000000","message":"Patch Set 56:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-12/319379/56/check/tempest-dsvm-full-nfv-xenial/39a2220 : FAILURE in 2h 10m 28s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-12/319379/56/check/tempest-dsvm-intel-nfv-xenial/35f388c : SUCCESS in 42m 38s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-12/319379/56/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/7e7c8ea : FAILURE in 1h 35m 32s","accounts_in_message":[],"_revision_number":56},{"id":"16aa6f58326d0863525a685ebb000b0529f96623","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-13 20:48:18.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.","accounts_in_message":[],"_revision_number":57},{"id":"d5405cea80f495c7fa8ba3b55618ae432954423b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-13 21:29:32.000000000","message":"Patch Set 57:\n\n* pci-test http://52.27.155.124/pci/319379/57 : SUCCESS","accounts_in_message":[],"_revision_number":57},{"id":"32524b5cd1ac3d0b0a238c333d09ff7c5739ebb5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-13 21:48:24.000000000","message":"Patch Set 57:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/57/check/check-dsvm-tempest-vz7-exe-minimal/0fcc866 : SUCCESS in 57m 18s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":57},{"id":"92569bc1aeac6e59af8df40d805dcbbfc5d2d489","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-13 21:53:48.000000000","message":"Patch Set 57: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/57/check/tempest-dsvm-full-xenial/48e76fe/ : FAILURE in 57m 27s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/57/check/grenade-dsvm-xenial/c3333ec/ : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"c52ee88379ac19ca99742c176aeeeb30dd8fe3a4","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-13 22:13:27.000000000","message":"Patch Set 57:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/57/ : FAILURE in 1h 08m 52s","accounts_in_message":[],"_revision_number":57},{"id":"2400b1397f5d59483c0838fe10b93391e4a61b52","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-13 22:30:07.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/57/check/gate-nova-docs-ubuntu-xenial/7d26e2e//doc/build/html/ : SUCCESS in 5m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-nova-pep8-ubuntu-xenial/954b3e1/ : SUCCESS in 3m 38s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-nova-python27-db-ubuntu-xenial/f32fcbc/ : SUCCESS in 10m 31s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/57/check/gate-nova-python35-db/d33b8b0/ : SUCCESS in 11m 07s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/638939e/ : FAILURE in 1h 38m 30s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-grenade-dsvm-neutron-ubuntu-xenial/487d950/ : SUCCESS in 58m 42s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-nova-tox-db-functional-ubuntu-xenial/96f310a/ : SUCCESS in 11m 06s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/7b3fe39/ : SUCCESS in 10m 04s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/d01acb9/ : SUCCESS in 39m 07s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/57/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/61eca65/ : SUCCESS in 33m 21s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-cells-ubuntu-xenial/82d47d5/ : FAILURE in 54m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0bbf0e3/ : FAILURE in 1h 02m 55s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/cedc943/ : FAILURE in 1h 17m 31s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f31c89d/ : SUCCESS in 1h 28m 16s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a54c405/ : FAILURE in 42m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/57/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ed05c97/ : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"d332b06fb980805c1b5bc0c8697d04b38178776d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-13 22:47:26.000000000","message":"Patch Set 57:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-13/319379/57/check/tempest-dsvm-full-nfv-xenial/5389c9e : FAILURE in 1h 54m 38s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-13/319379/57/check/tempest-dsvm-intel-nfv-xenial/070ac02 : SUCCESS in 45m 40s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-13/319379/57/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/5be5ed3 : FAILURE in 1h 42m 46s","accounts_in_message":[],"_revision_number":57},{"id":"41813a70759dd2b4cdd120dfb7a8f64e80f7096e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-13 22:48:06.000000000","message":"Patch Set 57:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/57/check/dsvm-tempest-neutron-network/5f47211 : FAILURE in 1h 48m 30s","accounts_in_message":[],"_revision_number":57},{"id":"2379ab920126e23dacc575611582539e928c6e2a","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-14 00:20:50.000000000","message":"Patch Set 57:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/57/check/dsvm-tempest-kvm/44a60b6 : FAILURE in 1h 04m 10s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/57/check/dsvm-tempest-exe/d783db0 : SUCCESS in 41m 25s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":57},{"id":"048b4ef186f5cd59343e8b64b15985b1e87be5b9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-14 00:27:26.000000000","message":"Patch Set 57:\n\nBuild failed. Comment recheck powervm to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/57/check/nova-pvm-dsvm-tempest-full/2f16582 : FAILURE in 3h 37m 22s","accounts_in_message":[],"_revision_number":57},{"id":"9dda0a3e851df21946542ea6d872c6a3a1ad5ca5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-14 02:15:07.000000000","message":"Patch Set 57:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/57/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/a8c9b19 : SUCCESS in 50m 37s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":57},{"id":"7b7f266d53c11c6823f67ec9097c85c0685e8ddc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-14 02:22:34.000000000","message":"Patch Set 57:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/57/check-nova/Nova-ML2-Sriov/c999b3a : SUCCESS in 44m 31s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/57/check-nova/Nova-MACVTAP-ML2-Sriov/4731117 : SUCCESS in 39m 36s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":57},{"id":"c91a8d8b120944a00aa11c7eb0e21e4908b4a7b4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-14 19:58:18.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.","accounts_in_message":[],"_revision_number":58},{"id":"4882050ae0daf0c216bb2227254b3caa4c29d73b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-14 20:33:45.000000000","message":"Patch Set 58:\n\n* pci-test http://52.27.155.124/pci/319379/58 : SUCCESS","accounts_in_message":[],"_revision_number":58},{"id":"5758e4e6acbc26fec0f9f3145d2e850fde60d5b3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-14 20:55:11.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/58/check/dsvm-tempest-kvm/23efbd7 : SUCCESS in 54m 32s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/58/check/dsvm-tempest-exe/8648454 : SUCCESS in 41m 39s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":58},{"id":"55b2249cc825930e17381a9c17058f2558d121ce","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-14 20:57:59.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/58/check/check-dsvm-tempest-vz7-exe-minimal/e3a0df8 : SUCCESS in 57m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":58},{"id":"962db22eb3b415d33a998f51601bef9732aafc44","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-14 21:10:09.000000000","message":"Patch Set 58: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/58/check/tempest-dsvm-full-xenial/964eb2b/ : FAILURE in 1h 06m 58s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/58/check/grenade-dsvm-xenial/6129e4d/ : SUCCESS in 44m 29s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"7547a7609dd90cf93977ef02583965429986ba77","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-14 21:17:35.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/58/check-nova/Nova-ML2-Sriov/b5c4abe : SUCCESS in 49m 03s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/58/check-nova/Nova-MACVTAP-ML2-Sriov/1a3992a : SUCCESS in 47m 49s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":58},{"id":"b327dfde65558dbaddbf9a22f18e20d046d8f6be","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-14 21:31:05.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/58/check/gate-nova-docs-ubuntu-xenial/ddabc55//doc/build/html/ : SUCCESS in 4m 05s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-nova-pep8-ubuntu-xenial/eab52b6/ : SUCCESS in 5m 46s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-nova-python27-db-ubuntu-xenial/3b188be/ : SUCCESS in 10m 50s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/58/check/gate-nova-python35-db/6360ae5/ : SUCCESS in 12m 08s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/816bade/ : SUCCESS in 1h 25m 21s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a713b7b/ : SUCCESS in 55m 28s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-nova-tox-db-functional-ubuntu-xenial/84f51fb/ : SUCCESS in 15m 13s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/a85ef8d/ : SUCCESS in 10m 02s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/1fcdf90/ : SUCCESS in 53m 03s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/58/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/8adc0e5/ : SUCCESS in 1h 10m 40s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-cells-ubuntu-xenial/a2419da/ : FAILURE in 46m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9d1adeb/ : FAILURE in 1h 18m 11s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/d3ef57d/ : FAILURE in 1h 23m 38s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/fef0ca8/ : SUCCESS in 1h 13m 52s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c25f7d1/ : FAILURE in 45m 48s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/58/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/9ed1d55/ : SUCCESS in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"431b994e013bbff9ebb4b5a3ca6d042d907df018","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-14 21:31:57.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/58/ : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":58},{"id":"3dc3413dbb10b3c761589e47c421c52bcf0536dd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-14 21:48:09.000000000","message":"Patch Set 58:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/58/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/f02ebf1 : FAILURE in 48m 43s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":58},{"id":"449dc721ba95e24a448e913e460f02b79dd52ca7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-14 22:08:43.000000000","message":"Patch Set 58:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/58/check/dsvm-tempest-neutron-network/cac491e : FAILURE in 2h 04m 32s","accounts_in_message":[],"_revision_number":58},{"id":"afe1ff8682156502d24c03bab8dfa12964c2ca82","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-14 22:08:52.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-14/319379/58/check/tempest-dsvm-full-nfv-xenial/9afd7b8 : FAILURE in 2h 06m 57s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-14/319379/58/check/tempest-dsvm-intel-nfv-xenial/edbb099 : SUCCESS in 41m 30s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-14/319379/58/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/a070589 : FAILURE in 1h 27m 01s","accounts_in_message":[],"_revision_number":58},{"id":"410c5f077fff0fd9b230160f8f4c701dbbcd74ed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-16 21:24:25.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"612e2d22435da4894b8b9ad9a1e634c445936830","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-16 21:52:36.000000000","message":"Patch Set 59:\n\n* pci-test http://52.27.155.124/pci/319379/59 : SUCCESS","accounts_in_message":[],"_revision_number":59},{"id":"6c3a0191ee9cf1e78838608e22b1bb74af1a2ff6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-16 22:21:11.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"97052d2358dbd5bb1ccb6ae06101f32f22a0a3b3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-16 22:44:29.000000000","message":"Patch Set 60:\n\n* pci-test http://52.27.155.124/pci/319379/60 : SUCCESS","accounts_in_message":[],"_revision_number":60},{"id":"869a355dedc008bb48aaeef3b7e30727369fa745","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-16 23:07:53.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"9bfbbd195819b97fcd9d1f7de771e4e18682c7eb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-16 23:36:31.000000000","message":"Patch Set 61:\n\n* pci-test http://52.27.155.124/pci/319379/61 : SUCCESS","accounts_in_message":[],"_revision_number":61},{"id":"5ae9e70feea490c634da90d84e23a45b039eb677","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-16 23:44:58.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"e3befb0fe5e508a724a2c7efba14533265c18021","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 00:08:33.000000000","message":"Patch Set 62: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"5f13bb9fac10c12e59a2ae567b2d7ab191ccb57d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-17 00:08:35.000000000","message":"Patch Set 62: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"064be4a00fb77fc48b45f70d9aa829f60f7f0e00","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-17 00:08:36.000000000","message":"Patch Set 62:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"8a86fa1f51094d991b63d1d1658dc168c12a106f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 00:08:38.000000000","message":"Patch Set 62:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"8f5c34b03b365866b192d78725b32bf326555a4c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 00:08:39.000000000","message":"Patch Set 62:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":62},{"id":"49c79ce2dabc53b6b0f98ca4e282ba6cabfe9271","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-17 00:08:42.000000000","message":"Patch Set 62:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"ab8b8eb831973e0c0c33b2ca08e425c695efbf29","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 00:08:47.000000000","message":"Patch Set 62:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":62},{"id":"5d63c1d36ece8bafa682dc7ffe0e62837797ec80","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-17 00:18:22.000000000","message":"Patch Set 62:\n\nrecheck dep updated","accounts_in_message":[],"_revision_number":62},{"id":"f51b7a043f272b5b853540b34ce76125a5809385","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-17 00:18:46.000000000","message":"Patch Set 62:\n\nrecheck gdi mriedem","accounts_in_message":[],"_revision_number":62},{"id":"e22490a8488f2523d9676c37cd9206896cb7bc90","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-17 00:32:53.000000000","message":"Patch Set 62:\n\n* pci-test http://52.27.155.124/pci/319379/62 : SUCCESS","accounts_in_message":[],"_revision_number":62},{"id":"c0b13010729eaf7c2a69f523dea2e23577822ce6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-17 01:24:49.000000000","message":"Patch Set 62:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/62/check/tempest-dsvm-full-xenial/d69986f/ : FAILURE in 1h 05m 47s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/62/check/grenade-dsvm-xenial/efdb0e6/ : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"9ea4e6a2261b9ff5454827261eac733429af70bf","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-17 01:26:34.000000000","message":"Patch Set 62:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/62/ : FAILURE in 1h 07m 16s","accounts_in_message":[],"_revision_number":62},{"id":"8b6b87c07b06b7af48735f9c6fab40b128fa1215","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 01:50:55.000000000","message":"Patch Set 62:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/62/check/gate-nova-docs-ubuntu-xenial/c55075c//doc/build/html/ : SUCCESS in 3m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-nova-pep8-ubuntu-xenial/169c0a6/ : FAILURE in 6m 05s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-nova-python27-db-ubuntu-xenial/ea3a553/ : FAILURE in 2m 50s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/62/check/gate-nova-python35-db/07cfe61/ : FAILURE in 5m 10s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/cbd0371/ : SUCCESS in 1h 28m 25s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-grenade-dsvm-neutron-ubuntu-xenial/cfe1f8b/ : SUCCESS in 31m 28s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-nova-tox-db-functional-ubuntu-xenial/1ebc505/ : FAILURE in 10m 06s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/d283379/ : FAILURE in 12m 34s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/12d9ce0/ : SUCCESS in 1h 02m 42s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/62/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/bd08ad1/ : SUCCESS in 34m 41s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-cells-ubuntu-xenial/6194f19/ : FAILURE in 1h 09m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/6c313fb/ : FAILURE in 1h 30m 43s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/8d4ca90/ : FAILURE in 27m 35s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/065e51f/ : FAILURE in 25m 52s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/62/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c178095/ : SUCCESS in 49m 06s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"0e7dbdf1090b3355b06db251db84693693627965","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-17 02:20:47.000000000","message":"Uploaded patch set 63.","accounts_in_message":[],"_revision_number":63},{"id":"b4060bd6872c2ba320940c2acd130758d91f0b3c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 02:28:27.000000000","message":"Patch Set 63: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":63},{"id":"c827455fa3800a40802e778ae8606d2bc5aad71b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-17 02:28:31.000000000","message":"Patch Set 63:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":63},{"id":"6c9d2e0733721a0539f11baf343314e6800849f6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 02:28:32.000000000","message":"Patch Set 63:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":63},{"id":"c094c6cf7752ca475eb75322d50f8256898f2953","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-17 02:28:36.000000000","message":"Patch Set 63:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":63},{"id":"f2246cacf0a89e64b64fa291120a11c67d425155","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 02:28:36.000000000","message":"Patch Set 63:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":63},{"id":"f6b417cce55d62c63758b4829341638cc296a2fe","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 02:28:41.000000000","message":"Patch Set 63:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":63},{"id":"552cab6b058d259e6695a86725589dd6c32c2df3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-17 02:31:34.000000000","message":"Patch Set 63:\n\nrecheck","accounts_in_message":[],"_revision_number":63},{"id":"d3bdef8e08a1e129d215efb149ec50e8e47052aa","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-17 02:42:32.000000000","message":"Patch Set 63:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/63/ : FAILURE in 10m 13s","accounts_in_message":[],"_revision_number":63},{"id":"555522c35df9490d1ac655d1f8ec982dc37a141d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-17 03:18:06.000000000","message":"Patch Set 63:\n\n* pci-test http://52.27.155.124/pci/319379/63 : SUCCESS","accounts_in_message":[],"_revision_number":63},{"id":"07f83a8129b9ac587d23a738280099434e6002e4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 03:29:31.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/63/check/check-dsvm-tempest-vz7-exe-minimal/3c82f8b : SUCCESS in 57m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":63},{"id":"90c1e25ff4f515c338a71c460d29e6e75385414b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-17 03:37:32.000000000","message":"Patch Set 63: Verified-1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/63/check/tempest-dsvm-full-xenial/8c8af8e/ : FAILURE in 1h 05m 17s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/63/check/grenade-dsvm-xenial/bef9391/ : SUCCESS in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"bc17f2679e43dde4af957406e23374a54e436de3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 04:07:46.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/63/check/gate-nova-docs-ubuntu-xenial/d762782//doc/build/html/ : SUCCESS in 2m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-pep8-ubuntu-xenial/7521e5e/ : SUCCESS in 3m 30s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-python27-db-ubuntu-xenial/6f2f837/ : FAILURE in 7m 45s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/63/check/gate-nova-python35-db/7af645b/ : FAILURE in 7m 10s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a0ccac0/ : FAILURE in 1h 34m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b021df0/ : SUCCESS in 53m 35s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-tox-db-functional-ubuntu-xenial/31a80d7/ : SUCCESS in 14m 17s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/bddfa65/ : SUCCESS in 14m 17s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/6bfe1af/ : SUCCESS in 36m 08s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/271c8e6/ : SUCCESS in 33m 06s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-cells-ubuntu-xenial/ece67ef/ : FAILURE in 37m 02s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/691d019/ : FAILURE in 1h 08m 15s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e02a63a/ : FAILURE in 26m 45s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/4bf230e/ : FAILURE in 27m 39s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ae3f63e/ : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"a65ca30cc1f08a1736eb16a3171af841df8890fa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 04:10:45.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-17/319379/63/check/tempest-dsvm-full-nfv-xenial/59e7dd3 : FAILURE in 1h 37m 44s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-17/319379/63/check/tempest-dsvm-intel-nfv-xenial/09030e4 : SUCCESS in 42m 18s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-17/319379/63/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/12abbb9 : FAILURE in 1h 24m 24s","accounts_in_message":[],"_revision_number":63},{"id":"3949c98f21b8fae48d88e4890658c7dfc6794540","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-17 07:05:35.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/63/check/dsvm-tempest-kvm/13b3d0f : SUCCESS in 58m 52s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/63/check/dsvm-tempest-exe/802aab7 : SUCCESS in 47m 06s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":63},{"id":"d1f60386425247cf0fe6bd123134071e9b5827a6","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-17 08:42:12.000000000","message":"Patch Set 62:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-pep8-ubuntu-xenial: unrecognized error\n- gate-nova-python27-db-ubuntu-xenial: unrecognized error\n- gate-nova-python35-db: unrecognized error\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: https://bugs.launchpad.net/bugs/1449136\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":62},{"id":"21dff280a5ad7cee15d13a8fb7d5fe796b7c5df6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-17 19:22:41.000000000","message":"Patch Set 63:\n\nrecheck should be good now","accounts_in_message":[],"_revision_number":63},{"id":"9ad53090f060a2b768c30d914e01d553e9a940e2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-17 20:34:41.000000000","message":"Patch Set 63:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/63/check/tempest-dsvm-full-xenial/af6d295/ : FAILURE in 1h 02m 59s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/63/check/grenade-dsvm-xenial/1eafd7b/ : SUCCESS in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"6116a70784860d1f51afb4e7faeccff285c7daac","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-17 20:42:10.000000000","message":"Patch Set 63:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/63/ : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":63},{"id":"1ff9ff9063a23062a05f88b22ec5a0775e12931c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 21:02:31.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/63/check/gate-nova-docs-ubuntu-xenial/8c47b17//doc/build/html/ : SUCCESS in 3m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-pep8-ubuntu-xenial/6a062f8/ : SUCCESS in 7m 29s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-python27-db-ubuntu-xenial/3b504c3/ : FAILURE in 7m 26s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/63/check/gate-nova-python35-db/181c900/ : FAILURE in 7m 34s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4abb1c1/ : SUCCESS in 1h 28m 31s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d27af75/ : SUCCESS in 50m 32s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-tox-db-functional-ubuntu-xenial/fe62c55/ : SUCCESS in 14m 39s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/2bed41f/ : SUCCESS in 11m 00s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/af427cf/ : SUCCESS in 1h 07m 35s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/63/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/7b6224a/ : SUCCESS in 35m 43s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-cells-ubuntu-xenial/c42c7d5/ : FAILURE in 40m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/fa9ee92/ : FAILURE in 1h 02m 57s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/279b1f7/ : FAILURE in 1h 04m 40s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3a166e0/ : SUCCESS in 27m 08s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/63/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/697a701/ : SUCCESS in 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"112254b10307e4d2ffa22df1991e8d1f45acd637","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-20 02:26:47.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"bd285f0636afa0f0417426513009f1d2b01b1026","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-20 02:53:51.000000000","message":"Patch Set 64:\n\n* pci-test http://52.27.155.124/pci/319379/64 : SUCCESS","accounts_in_message":[],"_revision_number":64},{"id":"12bd241a063589b8d06ad01809fa4c86ed92f802","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 03:02:37.000000000","message":"Patch Set 64: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"d836883440f6ad5e2109692b68ef46711cbd6774","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-20 03:02:38.000000000","message":"Patch Set 64: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"bf483c3cf13b0bdca5d1a10d02312bd656a69065","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-20 03:02:41.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"b46f6edec64055c127f8c3d5e2df41ca315fe924","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-20 03:02:42.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"43cf9f42bbbad325031e57497c7674f5e141cf3e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-20 03:02:44.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"4e26b2e9fb4d4f8234ff5fe4350cacd4cae53b15","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-20 03:02:50.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"daa8b45025af0323fb4eac52d4b98d9ad4f50307","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-20 03:03:34.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":64},{"id":"1ebe08684a94294534128ff57847ac3f299b0d31","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-20 03:03:54.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":64},{"id":"6f8c1cd2f5982f1135d9f8f88ac34f0c7a5ad6d4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-20 03:04:02.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":64},{"id":"9c19f5fbea16aa9e9bb171152e67435eebedc84a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-20 03:05:35.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":64},{"id":"75224d744d63b95991977b6c2b9f67b1777b3da3","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-20 06:11:35.000000000","message":"Patch Set 64:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":64},{"id":"17877623100b8eee22ea0cf6c4213f009ff98d08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-20 07:12:11.000000000","message":"Patch Set 64:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/tempest-dsvm-full-xenial/75f5c10/ : FAILURE in 59m 42s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/grenade-dsvm-xenial/748a3fc/ : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"be52cdbc6b8eb77a272b40b31e3c9a965684544f","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-20 07:28:53.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/64/ : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":64},{"id":"85d6cda4a7ec8ac69bc474c280cfd51ce1c26baf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 07:37:08.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/64/check/gate-nova-docs-ubuntu-xenial/25f6417//doc/build/html/ : SUCCESS in 3m 00s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-pep8-ubuntu-xenial/512be9c/ : SUCCESS in 3m 34s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-python27-db-ubuntu-xenial/576a2bb/ : SUCCESS in 10m 52s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/64/check/gate-nova-python35-db/9acefaa/ : SUCCESS in 7m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/839cd04/ : SUCCESS in 1h 02m 21s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d741647/ : SUCCESS in 37m 19s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-ubuntu-xenial/27c69e7/ : SUCCESS in 11m 10s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/4fa04c9/ : SUCCESS in 15m 44s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/3d7d90e/ : SUCCESS in 36m 32s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/ce4ed1e/ : SUCCESS in 33m 49s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-cells-ubuntu-xenial/f051790/ : SUCCESS in 48m 08s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b0273f4/ : FAILURE in 1h 23m 28s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/bd4b3a6/ : FAILURE in 55m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c0f62cb/ : SUCCESS in 24m 02s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c1c8a7a/ : SUCCESS in 40m 33s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"7a87d70221a4f0a8a7f9d51f6cc96731a41af5b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-20 08:50:02.000000000","message":"Patch Set 64:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-64 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":64},{"id":"5013753854b096d075ab933b7e60e3385f2390f8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-20 14:43:24.000000000","message":"Patch Set 64:\n\nrecheck dep updated","accounts_in_message":[],"_revision_number":64},{"id":"3c471a17d22981d4565e1c3a01e8e9b6e23e2617","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-20 15:18:25.000000000","message":"Patch Set 64:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/tempest-dsvm-full-xenial/2452402/ : FAILURE in 29m 29s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/grenade-dsvm-xenial/cebc627/ : FAILURE in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"670fc8386c6920b537ece193a8095844732e9e18","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-20 15:20:22.000000000","message":"Patch Set 64:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"c75b324f14b8fa991afc2a63906e4ef3b52ffb33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 15:20:31.000000000","message":"Patch Set 64: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":64},{"id":"70f76cf54fef082f54e58aacc2b3197c98836a88","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-01-20 15:22:01.000000000","message":"Patch Set 64:\n\nrecheck dep update","accounts_in_message":[],"_revision_number":64},{"id":"59d7379bcc2c8ed15c45205fa221550f5a88bff9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-20 16:21:47.000000000","message":"Patch Set 64:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/tempest-dsvm-full-xenial/971f7c3/ : FAILURE in 58m 34s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/grenade-dsvm-xenial/6331737/ : FAILURE in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"0e3d64415f56a847871df95df14f7c2ed91bc600","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-20 17:00:32.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/64/ : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":64},{"id":"cdb3c5b8856bd41bc535a078cd528ebda9c3f20f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-20 17:19:01.000000000","message":"Patch Set 64:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/64/check/gate-nova-docs-ubuntu-xenial/ab1cba8//doc/build/html/ : SUCCESS in 2m 46s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-pep8-ubuntu-xenial/05acba8/ : SUCCESS in 4m 03s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-python27-db-ubuntu-xenial/d76eb35/ : SUCCESS in 8m 02s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/64/check/gate-nova-python35-db/8290f2e/ : SUCCESS in 8m 30s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8ff6202/ : FAILURE in 1h 02m 53s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-ubuntu-xenial/49c8367/ : FAILURE in 27m 26s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-ubuntu-xenial/9ed68db/ : SUCCESS in 11m 32s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/1b17b2d/ : SUCCESS in 11m 17s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/abb79fb/ : FAILURE in 1h 01m 09s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/1781d6f/ : FAILURE in 57m 02s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-cells-ubuntu-xenial/49e452c/ : SUCCESS in 37m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ea96c7e/ : SUCCESS in 1h 04m 29s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/ecb406b/ : SUCCESS in 1h 49m 47s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8654df4/ : SUCCESS in 25m 52s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f8e1dcd/ : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"16a04f9f9c164b4d39562edd33905078800067a0","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-20 23:57:15.000000000","message":"Patch Set 64:\n\nrecheck dep updated","accounts_in_message":[],"_revision_number":64},{"id":"efea842b7e2462c6916dd43e9da6a94e4d6bafe1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-21 01:00:55.000000000","message":"Patch Set 64:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/64/check/gate-nova-docs-ubuntu-xenial/0639b11//doc/build/html/ : SUCCESS in 3m 13s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-pep8-ubuntu-xenial/cfc9392/ : SUCCESS in 6m 23s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-python27-db-ubuntu-xenial/86bddbf/ : SUCCESS in 12m 11s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/64/check/gate-nova-python35-db/b58ec96/ : SUCCESS in 11m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/fb4bf84/ : FAILURE in 58m 19s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ef2f245/ : SUCCESS in 33m 04s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-ubuntu-xenial/493493d/ : SUCCESS in 15m 49s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/3620bf7/ : SUCCESS in 16m 43s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ba2b066/ : SUCCESS in 37m 19s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/64/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/667ee58/ : FAILURE in 31m 13s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-cells-ubuntu-xenial/10be83a/ : FAILURE in 55m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7391ff1/ : FAILURE in 1h 01m 46s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/b260f7b/ : FAILURE in 37m 40s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/674a5f8/ : FAILURE in 37m 17s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/64/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/703554c/ : SUCCESS in 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"a13a4a65aba3bbc66be1ad37c44ff94cd749da8b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-21 01:08:02.000000000","message":"Patch Set 64:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/64/ : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":64},{"id":"7d988b7ce3cc57b9ecb777406160d54fb535a13e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-21 01:17:39.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/tempest-dsvm-full-xenial/21f1c42/ : SUCCESS in 1h 19m 17s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/64/check/grenade-dsvm-xenial/d56a220/ : SUCCESS in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"3f7f4185ecb5eafd2967f0337c4e46c30c2562ee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-21 03:04:46.000000000","message":"Patch Set 56:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/56/2096/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":56},{"id":"0c8d64c2122f6bb75daf1666855fc46df8970cef","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-22 00:24:58.000000000","message":"Uploaded patch set 66.","accounts_in_message":[],"_revision_number":66},{"id":"a947c3c6bb704739057f9c7055b5483c807a3aed","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-22 00:45:40.000000000","message":"Patch Set 66:\n\n* pci-test http://52.27.155.124/pci/319379/66 : SUCCESS","accounts_in_message":[],"_revision_number":66},{"id":"54dfd6d300586133c47cf826ebe0d0e00c6ef02e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-22 01:15:06.000000000","message":"Patch Set 66:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/66/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/1855c0a : FAILURE in 44m 46s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":66},{"id":"fe1edb270367164a0b0b505a200caddbf184cddf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-22 01:17:42.000000000","message":"Patch Set 66:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/66/check-nova/Nova-ML2-Sriov/eb60b66 : SUCCESS in 42m 03s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/66/check-nova/Nova-MACVTAP-ML2-Sriov/72f4427 : SUCCESS in 43m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":66},{"id":"55dd6b0f09affe7aeb5104ab78e2135600a2bff3","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-22 01:20:21.000000000","message":"Patch Set 66:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/66/ : FAILURE in 54m 17s","accounts_in_message":[],"_revision_number":66},{"id":"297e350d6a71f95ba0192e54a842fd003a20cc55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-22 01:32:00.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/66/check/gate-nova-docs-ubuntu-xenial/63ed19f//doc/build/html/ : SUCCESS in 2m 12s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-nova-pep8-ubuntu-xenial/b057665/ : SUCCESS in 3m 22s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-nova-python27-db-ubuntu-xenial/81576d6/ : SUCCESS in 9m 42s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/66/check/gate-nova-python35-db/47f1be1/ : SUCCESS in 6m 29s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4e242f6/ : FAILURE in 1h 05m 34s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-grenade-dsvm-neutron-ubuntu-xenial/cc916b9/ : SUCCESS in 30m 03s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-nova-tox-db-functional-ubuntu-xenial/6b95307/ : SUCCESS in 9m 42s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/fd09ee1/ : SUCCESS in 14m 36s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/22e998a/ : SUCCESS in 1h 00m 45s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/66/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/7714532/ : FAILURE in 30m 03s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-cells-ubuntu-xenial/f52c65f/ : SUCCESS in 33m 35s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8e552fe/ : SUCCESS in 1h 02m 24s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/50c0c12/ : FAILURE in 39m 16s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/48b6f12/ : FAILURE in 40m 23s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/66/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/db99cd9/ : SUCCESS in 26m 49s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"9947c5c202e216a0df0e811cae524669bbdf7f2f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-22 01:46:00.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"5bc109905f53a72a23d5d9a02df6ad5cb801d9c2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-22 02:11:45.000000000","message":"Patch Set 67:\n\n* pci-test http://52.27.155.124/pci/319379/67 : SUCCESS","accounts_in_message":[],"_revision_number":67},{"id":"61d7bf6ef246f4c6029d30a138427deb82d7b3f7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-22 02:38:19.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/67/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/e9cd17d : SUCCESS in 47m 09s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":67},{"id":"bd2034b9881af2b002a4b73bfa90d0204644be89","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-22 02:40:40.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/67/check-nova/Nova-ML2-Sriov/0570227 : SUCCESS in 45m 25s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/67/check-nova/Nova-MACVTAP-ML2-Sriov/28ea489 : SUCCESS in 46m 03s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":67},{"id":"f4a10995d9e021ca679dc88032a3e5c9a1c210b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-22 02:50:16.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/67/check/gate-nova-docs-ubuntu-xenial/207a330//doc/build/html/ : SUCCESS in 2m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-nova-pep8-ubuntu-xenial/5997439/ : SUCCESS in 3m 24s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-nova-python27-db-ubuntu-xenial/3408da3/ : SUCCESS in 9m 46s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/67/check/gate-nova-python35-db/31ecf69/ : SUCCESS in 7m 08s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ab0a72d/ : SUCCESS in 1h 02m 43s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-grenade-dsvm-neutron-ubuntu-xenial/eb0f4cd/ : SUCCESS in 31m 36s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-nova-tox-db-functional-ubuntu-xenial/facd130/ : SUCCESS in 10m 20s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/00ae46e/ : SUCCESS in 9m 32s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/aade249/ : SUCCESS in 36m 33s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/67/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/d966f94/ : FAILURE in 31m 53s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-cells-ubuntu-xenial/c367062/ : SUCCESS in 34m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/42233ce/ : SUCCESS in 59m 41s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/674af37/ : FAILURE in 24m 00s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/c01053b/ : FAILURE in 36m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/67/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/738c66c/ : SUCCESS in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"041773600a144b243b6b2b28581a857233917968","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-22 02:51:33.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/67/check/check-dsvm-tempest-vz7-exe-minimal/7ae9ba7 : SUCCESS in 1h 04m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":67},{"id":"ffdff1136fba6a9d966c2fad5169db042a7a868d","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-22 02:57:50.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/67/ : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":67},{"id":"93ad75ad65ef3b9090be721a37e9634af817109b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-22 03:40:47.000000000","message":"Patch Set 67:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/67/check/dsvm-tempest-neutron-network/ab505a9 : FAILURE in 1h 51m 55s","accounts_in_message":[],"_revision_number":67},{"id":"b6c1bc3f037c3c68af6261c01fb50c70ff1e4fc4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-22 03:59:07.000000000","message":"Patch Set 67:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-22/319379/67/check/tempest-dsvm-full-nfv-xenial/3d98cb1 : SUCCESS in 1h 59m 07s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-22/319379/67/check/tempest-dsvm-intel-nfv-xenial/1a35ece : SUCCESS in 1h 02m 07s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-22/319379/67/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/dadcb36 : FAILURE in 1h 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"076c2c11878ca626a675ca833f2097d178393d80","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-22 09:55:57.000000000","message":"Patch Set 67:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/67/check/dsvm-tempest-kvm/c65d5e5 : SUCCESS in 58m 14s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/67/check/dsvm-tempest-exe/6b72990 : SUCCESS in 42m 09s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":67},{"id":"7d814e6d8a4c19f5c89b7225eb30ebc1fce73914","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-23 16:35:28.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.","accounts_in_message":[],"_revision_number":68},{"id":"563cf079fa04ea608af7a1a22c67335dbf0231ad","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-23 17:03:07.000000000","message":"Patch Set 68:\n\n* pci-test http://52.27.155.124/pci/319379/68 : SUCCESS","accounts_in_message":[],"_revision_number":68},{"id":"b5f3f204810f00f68c6b5aad0225319aef438e0a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-23 17:26:33.000000000","message":"Uploaded patch set 69: Commit message was updated.","accounts_in_message":[],"_revision_number":69},{"id":"4fa7adc6b5fad51941e86c40a7e8abe1b5a507fc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 17:47:11.000000000","message":"Patch Set 69: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":69},{"id":"ef3ad773ec94ae08b74575c7e0c4ad74e01fbbd1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:13.000000000","message":"Patch Set 69: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":69},{"id":"8be7c442dadc24acdfe293588f631996a7adeb3f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:14.000000000","message":"Patch Set 69:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial tempest-dsvm-full-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"b12fc372b166cbaafd396457cdaeafa7cf054c4f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:16.000000000","message":"Patch Set 69:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":69},{"id":"4ab9c494e7f0e50b195cad4eb6fd6cc5ed045f0b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-23 17:47:16.000000000","message":"Patch Set 69:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":69},{"id":"48b3d3bc6031e08c52dac6f9e818afb8b4ee1e89","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:21.000000000","message":"Patch Set 69:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":69},{"id":"0098c089a84989c6b38ec4f83009ac03de860203","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:24.000000000","message":"Patch Set 69:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":69},{"id":"fd145fd1cfcf70d16c1fb05bb47e30893e2a6792","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:25.000000000","message":"Patch Set 69:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":69},{"id":"a5573395e4d0c46936c3b5c4836f451407b8065d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:39.000000000","message":"Patch Set 69:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":69},{"id":"3deabc5b32b78a78b7e98ebf9398757564e1ce43","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-23 17:47:40.000000000","message":"Patch Set 69:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":69},{"id":"45ebb413550bfe2bebd2d094e22e02c8d9803fdf","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-23 17:58:27.000000000","message":"Patch Set 69:\n\n* pci-test http://52.27.155.124/pci/319379/69 : SUCCESS","accounts_in_message":[],"_revision_number":69},{"id":"f1a258e904876b7c38ac7cd91ed93ec97d58045c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-23 17:59:10.000000000","message":"Patch Set 69:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":69},{"id":"d64350f2ee6979596725b33f32fbbf352fd29e6b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-23 18:17:25.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.","accounts_in_message":[],"_revision_number":70},{"id":"e264890506882965dcd2d68e0b471551f55e4a62","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-23 19:01:27.000000000","message":"Patch Set 70:\n\n* pci-test http://52.27.155.124/pci/319379/70 : SUCCESS","accounts_in_message":[],"_revision_number":70},{"id":"ccc429da972c2cbf4924cff36fa6ec107ae62587","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-23 19:26:33.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/70/check/tempest-dsvm-full-xenial/ffab89c/ : SUCCESS in 1h 07m 21s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/70/check/grenade-dsvm-xenial/f87acb0/ : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"342ac3577539321dcb9c59694988e3cd68047d07","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-23 19:30:14.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/70/ : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":70},{"id":"5cb2ec306e38f34b9c0639c0d8b436aa0cab5c3e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 19:31:47.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/70/check/check-dsvm-tempest-vz7-exe-minimal/b6fba11 : SUCCESS in 58m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":70},{"id":"faf124f5007b136b1626f932073d92901437ba06","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-23 19:47:15.000000000","message":"Patch Set 70:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/70/check/nova-pvm-dsvm-tempest-full/2c0e9e8 : FAILURE in 1h 28m 35s","accounts_in_message":[],"_revision_number":70},{"id":"98e2069575c5260564e73a38aa8ac27d17bc8f99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 19:59:34.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/70/check/gate-nova-docs-ubuntu-xenial/cd68894//doc/build/html/ : SUCCESS in 4m 23s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-pep8-ubuntu-xenial/4fa3147/ : SUCCESS in 5m 47s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-python27-db-ubuntu-xenial/3f09981/ : SUCCESS in 10m 45s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/70/check/gate-nova-python35-db/8e0f61d/ : SUCCESS in 10m 35s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9ff36ef/ : SUCCESS in 1h 12m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-ubuntu-xenial/914626b/ : FAILURE in 12m 23s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-tox-db-functional-ubuntu-xenial/a4ed64a/ : SUCCESS in 18m 45s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/3344ecb/ : SUCCESS in 15m 24s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/4aa0ae7/ : FAILURE in 21m 11s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/1bdd525/ : FAILURE in 11m 41s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/1076ab5/ : SUCCESS in 1h 00m 26s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-cells-ubuntu-xenial/7ef32e8/ : SUCCESS in 54m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4361352/ : FAILURE in 1h 25m 44s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/acde67e/ : SUCCESS in 1h 02m 19s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/de5505a/ : SUCCESS in 24m 37s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/21f982d/ : SUCCESS in 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"b35fa9749062b2f9fcc627b851c1245083d81dd7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 20:07:01.000000000","message":"Patch Set 70:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-23/319379/70/check/tempest-dsvm-full-nfv-xenial/1f6a28c : SUCCESS in 1h 41m 00s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-23/319379/70/check/tempest-dsvm-intel-nfv-xenial/5c8df31 : SUCCESS in 44m 55s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-23/319379/70/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/f0a29fc : FAILURE in 1h 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"ba36524e6e53b85ec51a38b2cf94da9d49a97c6d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-23 20:53:21.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/70/check/dsvm-tempest-neutron-network/b80faf9 : SUCCESS in 2h 10m 06s","accounts_in_message":[],"_revision_number":70},{"id":"a2adca9ef0316c2f95c220e4f7760c294f75b227","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-23 23:17:03.000000000","message":"Patch Set 70:\n\nrecheck infra bug","accounts_in_message":[],"_revision_number":70},{"id":"6eb989101dd1b032c1458a2bcfa40e64553cf4e3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-23 23:38:53.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/70/check/dsvm-tempest-kvm/bb5be5c : SUCCESS in 57m 29s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/70/check/dsvm-tempest-exe/c0eeae2 : SUCCESS in 42m 09s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":70},{"id":"65abde097b2e845fbaf2e11b961393a9d8472494","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-24 00:31:25.000000000","message":"Patch Set 70:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/70/check/tempest-dsvm-full-xenial/aa84dea/ : SUCCESS in 1h 13m 18s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/70/check/grenade-dsvm-xenial/3320f19/ : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"8d741c67a4be77db064f10339e77147bd6be5b03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 00:46:34.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/70/check/gate-nova-docs-ubuntu-xenial/e03c2b3//doc/build/html/ : SUCCESS in 2m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-pep8-ubuntu-xenial/b51b9f0/ : SUCCESS in 4m 34s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-python27-db-ubuntu-xenial/5900821/ : SUCCESS in 7m 02s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/70/check/gate-nova-python35-db/07d4c77/ : SUCCESS in 7m 11s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9f09bba/ : SUCCESS in 1h 18m 46s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-ubuntu-xenial/433bf89/ : SUCCESS in 51m 44s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-tox-db-functional-ubuntu-xenial/4ceac41/ : SUCCESS in 13m 53s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/1196f55/ : FAILURE in 14m 14s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b911e31/ : SUCCESS in 36m 24s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/70/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a809323/ : FAILURE in 33m 04s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/7a1f30d/ : FAILURE in 1h 26m 14s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-cells-ubuntu-xenial/7978133/ : SUCCESS in 47m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b24bc66/ : FAILURE in 1h 17m 37s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/4e65173/ : FAILURE in 1h 02m 42s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/68903fe/ : SUCCESS in 24m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/70/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/14bbbd6/ : SUCCESS in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"5889e54ddd1889a700558699e2745eba1987e1d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 00:56:28.000000000","message":"Uploaded patch set 71: Commit message was updated.","accounts_in_message":[],"_revision_number":71},{"id":"599d203b3d5dd1ab99e7e0da796ca5cf2fceee12","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 01:20:32.000000000","message":"Patch Set 71:\n\n* pci-test http://52.27.155.124/pci/319379/71 : FAILURE","accounts_in_message":[],"_revision_number":71},{"id":"a1d2d989e2fa37f8e9d8d71a9b8c531345f05ba1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-24 01:57:56.000000000","message":"Patch Set 71: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/71/check/tempest-dsvm-full-xenial/d356d79/ : FAILURE in 1h 00m 21s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/71/check/grenade-dsvm-xenial/51d358c/ : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"416af4bdd05ba2e4162e7c85eec6077f0cef8f76","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 01:59:44.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/71/check/check-dsvm-tempest-vz7-exe-minimal/76f3b30 : SUCCESS in 1h 01m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":71},{"id":"16a4bccee6a35f92df9fb200da28d9d2ad17087c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-24 02:05:43.000000000","message":"Patch Set 70:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-70 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":70},{"id":"e3c764068f9e032dd083436f61db73ec585c33c4","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-24 02:09:07.000000000","message":"Patch Set 71:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/71/check/nova-pvm-dsvm-tempest-full/872908d : FAILURE in 1h 11m 32s","accounts_in_message":[],"_revision_number":71},{"id":"bda096729dfd7cddc3a5752b48060622872b7912","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-24 02:12:21.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/71/ : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":71},{"id":"ebcd88f6e81c12c388785caaa01629bf8d320fdc","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-24 02:15:58.000000000","message":"Patch Set 71: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":71},{"id":"31fdbb3085d65956344f3b972c858b41e3753683","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 02:27:20.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/71/check/gate-nova-docs-ubuntu-xenial/2edbce2//doc/build/html/ : SUCCESS in 2m 53s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-nova-pep8-ubuntu-xenial/4b88608/ : SUCCESS in 3m 36s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-nova-python27-db-ubuntu-xenial/d5b7cd0/ : SUCCESS in 7m 06s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/71/check/gate-nova-python35-db/a967d58/ : SUCCESS in 10m 49s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/97a6d8c/ : SUCCESS in 1h 02m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-grenade-dsvm-neutron-ubuntu-xenial/f6a5caa/ : SUCCESS in 30m 34s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-nova-tox-db-functional-ubuntu-xenial/b110585/ : SUCCESS in 10m 58s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/e21cd42/ : SUCCESS in 15m 25s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/636d097/ : SUCCESS in 35m 15s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/71/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/045ea8c/ : FAILURE in 30m 28s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/a69695d/ : FAILURE in 1h 13m 18s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-cells-ubuntu-xenial/6e5906a/ : SUCCESS in 57m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8479fca/ : SUCCESS in 1h 26m 42s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/cdb61af/ : SUCCESS in 59m 31s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d2b2aa4/ : SUCCESS in 22m 57s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/71/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8c36668/ : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"9770dcb51a4473a0ab17739fe00070d5160491f2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 02:49:44.000000000","message":"Patch Set 71:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-24/319379/71/check/tempest-dsvm-full-nfv-xenial/dbf023b : SUCCESS in 1h 45m 34s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-24/319379/71/check/tempest-dsvm-intel-nfv-xenial/4a79e0a : SUCCESS in 41m 16s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-24/319379/71/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/c072ad1 : FAILURE in 1h 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"4b39100f4a66bfff84956a96226ce9c5bf4945c9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-24 03:19:29.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/71/check/dsvm-tempest-neutron-network/aea44ef : SUCCESS in 2h 15m 42s","accounts_in_message":[],"_revision_number":71},{"id":"f63ef5cc77fd0c1378cf8b93a06b6b614af0b8c9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-24 03:45:53.000000000","message":"Patch Set 71:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-71 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":71},{"id":"590a5e92123bab318f3471335fab72593e37b5cc","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 06:28:06.000000000","message":"Patch Set 71:\n\n* pci-test http://52.27.155.124/pci/319379/71 : SUCCESS","accounts_in_message":[],"_revision_number":71},{"id":"ff888f23f0f979f3d5d71fad91b59daee123ddcf","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 06:42:10.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/71/check/dsvm-tempest-kvm/b33c6af : SUCCESS in 55m 30s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/71/check/dsvm-tempest-exe/9ec156a : SUCCESS in 42m 43s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":71},{"id":"35692210e33a6cff78f4c57b4c99a913afc9fe4d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-24 07:06:36.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/71/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/578e276 : ABORTED in 0s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":71},{"id":"96ea50154c460a449fd6af36b00dbd2cf91f1b63","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-24 09:52:55.000000000","message":"Patch Set 71:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/71/check-nova/Nova-ML2-Sriov/71a0e6f : SUCCESS in 50m 23s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/71/check-nova/Nova-MACVTAP-ML2-Sriov/ebcce10 : SUCCESS in 55m 39s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":71},{"id":"ecc18fd1a8b488c2f3234c85863483899070b10f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 16:04:44.000000000","message":"Patch Set 71:\n\n(4 comments)","accounts_in_message":[],"_revision_number":71},{"id":"5bc1686f3b14c442bb9f1a59cb79a17f12495105","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 16:07:32.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"fe8404e2a82807fbb98f6b5160d32e5b598b5535","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 16:36:47.000000000","message":"Uploaded patch set 73.","accounts_in_message":[],"_revision_number":73},{"id":"9f1cb5283ea6a1707d2a89388421f0c8b3a4fdb0","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-24 16:57:33.000000000","message":"Patch Set 73: Code-Review+2\n\nLooks good to me now","accounts_in_message":[],"_revision_number":73},{"id":"ab767a5ba5fd42efeca1f48e587739f2fcde32e7","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-01-24 17:11:50.000000000","message":"Patch Set 73:\n\n(2 comments)","accounts_in_message":[],"_revision_number":73},{"id":"023ecbf3ba32ed07f4fe0aa5fb3a13a6a8da762b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-24 17:13:43.000000000","message":"Patch Set 73: -Code-Review\n\nWe\u0027re checking on the change in nova/tests/functional/integrated_helpers.py. It seems like it shouldn\u0027t be needed as we should get the same behavior even after moving instance create to conductor, via build request.","accounts_in_message":[],"_revision_number":73},{"id":"50acb4a14d44fbb8339f35f2ab5b89031006a6c7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 17:30:00.000000000","message":"Patch Set 73:\n\n* pci-test http://52.27.155.124/pci/319379/73 : SUCCESS","accounts_in_message":[],"_revision_number":73},{"id":"aa3a73a509a5902c2612ea25ead03104faeaa7bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 17:36:36.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"e4f45023db27e47e15ffccce52dff94411315e8a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 17:52:42.000000000","message":"Uploaded patch set 75: Patch Set 74 was rebased.","accounts_in_message":[],"_revision_number":75},{"id":"2f8a2b3b861273a4718a4edcdb053bfd6017c569","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 18:57:46.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/75/check/gate-nova-docs-ubuntu-xenial/7200834//doc/build/html/ : SUCCESS in 2m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-nova-pep8-ubuntu-xenial/d6f2e42/ : SUCCESS in 5m 31s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-nova-python27-db-ubuntu-xenial/19c9966/ : SUCCESS in 7m 28s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/75/check/gate-nova-python35-db/2fdd1cf/ : SUCCESS in 8m 09s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3e1f1da/ : SUCCESS in 54m 41s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-grenade-dsvm-neutron-ubuntu-xenial/630fb56/ : SUCCESS in 32m 29s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-nova-tox-db-functional-ubuntu-xenial/671b8ca/ : SUCCESS in 10m 40s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/fc6ca2c/ : SUCCESS in 10m 48s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e2329e5/ : SUCCESS in 37m 31s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/75/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2e56e28/ : FAILURE in 47m 13s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/8312fc6/ : SUCCESS in 52m 53s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-cells-ubuntu-xenial/5ef44d2/ : SUCCESS in 33m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/2bc1b8c/ : SUCCESS in 1h 00m 11s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/fcc2d7b/ : SUCCESS in 59m 15s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3754e15/ : SUCCESS in 24m 24s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/75/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4879424/ : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"c7c8dd2dbc29f69db154c03cd39053d4f1e57086","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-24 19:02:44.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/75/check/tempest-dsvm-full-xenial/9e81e3d/ : SUCCESS in 1h 05m 10s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/75/check/grenade-dsvm-xenial/f3e2889/ : SUCCESS in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"900066c15a2632c145a2b518ae28a9c8e2ac6e8d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 19:13:12.000000000","message":"Patch Set 75:\n\n* pci-test http://52.27.155.124/pci/319379/75 : SUCCESS","accounts_in_message":[],"_revision_number":75},{"id":"0ba5ae5a6c856e9b207e9c30283c1e117803f009","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-24 19:13:51.000000000","message":"Patch Set 75:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/75/check/nova-pvm-dsvm-tempest-full/6b225ee : FAILURE in 1h 18m 49s","accounts_in_message":[],"_revision_number":75},{"id":"29b6f7bc0674accdf51d6c6e29c2a9d1672fd3ed","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 19:19:48.000000000","message":"Patch Set 75:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":75},{"id":"1aa5e916a2051d2833de09b6fb74d44d82c71f3e","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-24 19:19:53.000000000","message":"Patch Set 75:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":75},{"id":"1cfd598e331c292068ff2458f4bb882603461eef","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 19:19:57.000000000","message":"Patch Set 75:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial tempest-dsvm-full-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"6ac617e7003741928caab241d66dd19ade67f63a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-24 19:20:06.000000000","message":"Patch Set 75:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":75},{"id":"7ffef278ac3adf146209eb97d8c5eb77f36588b4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 19:20:19.000000000","message":"Patch Set 75:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":75},{"id":"93470fc6f3708e2227d8a3fbc37eb66870b1dedb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 19:20:46.000000000","message":"Uploaded patch set 76.","accounts_in_message":[],"_revision_number":76},{"id":"4618e3b00c436cd0493ffc02d0182afb759e8426","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-24 19:20:59.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode Nova-ML2-Sriov-Multinode : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":75},{"id":"37a0568a6d5b5d6433383860679973a9de2569a3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-24 19:21:51.000000000","message":"Patch Set 75:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":75},{"id":"5ffce2f9c0a1fd5b62498cc16174267a85b7d0fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-24 20:24:04.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/76/check/tempest-dsvm-full-xenial/32cff57/ : SUCCESS in 1h 00m 49s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/76/check/grenade-dsvm-xenial/616afb2/ : SUCCESS in 1h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"57534b62f81775c3d0dd9f7412420aebb0805ce5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 20:26:06.000000000","message":"Patch Set 76:\n\n* pci-test http://52.27.155.124/pci/319379/76 : SUCCESS","accounts_in_message":[],"_revision_number":76},{"id":"5cbd9f50802dea98ca5d3d3f60d329a00028a7cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 20:28:59.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/76/check/gate-nova-docs-ubuntu-xenial/c3e8f00//doc/build/html/ : SUCCESS in 3m 41s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-nova-pep8-ubuntu-xenial/9afd1a0/ : SUCCESS in 5m 47s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-nova-python27-db-ubuntu-xenial/4855346/ : FAILURE in 7m 17s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/76/check/gate-nova-python35-db/e2084de/ : FAILURE in 6m 51s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/64834e8/ : SUCCESS in 57m 37s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-grenade-dsvm-neutron-ubuntu-xenial/51c025c/ : SUCCESS in 32m 16s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-nova-tox-db-functional-ubuntu-xenial/b314c23/ : SUCCESS in 14m 09s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/be614f7/ : SUCCESS in 14m 35s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/4d75bd1/ : SUCCESS in 35m 35s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/76/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/755b391/ : FAILURE in 32m 26s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/85634b8/ : FAILURE in 48m 27s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-cells-ubuntu-xenial/30401a9/ : SUCCESS in 37m 10s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/273a94b/ : FAILURE in 59m 00s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/8fbfe54/ : FAILURE in 1h 03m 10s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/7c4afd8/ : SUCCESS in 40m 07s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/76/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4412640/ : SUCCESS in 27m 11s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"7834fd355e73a4ac799ebaf6afac255fb5241956","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-24 20:41:41.000000000","message":"Patch Set 76:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/76/check/nova-pvm-dsvm-tempest-full/9b6f1ca : FAILURE in 1h 19m 42s","accounts_in_message":[],"_revision_number":76},{"id":"4cc66c615f85851c44a7b92a366632b319c1e770","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 21:01:32.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"682a51c87dd4135b4c1e719a56fd0da5fe16d8c2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-24 21:31:58.000000000","message":"Patch Set 77:\n\n* pci-test http://52.27.155.124/pci/319379/77 : SUCCESS","accounts_in_message":[],"_revision_number":77},{"id":"7528db8dcccae9d39b1766927a6ecfa6ac1e92eb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-24 22:01:45.000000000","message":"Patch Set 77: Code-Review-1\n\n(1 comment)\n\nLooks like one small thing was missed with the _validate_bdm removal","accounts_in_message":[],"_revision_number":77},{"id":"a33c0d93e1a9f776ff50db61e45d175b3fba63ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-24 22:08:10.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/77/check/tempest-dsvm-full-xenial/9f42b16/ : FAILURE in 1h 04m 30s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/77/check/grenade-dsvm-xenial/99e5905/ : SUCCESS in 56m 04s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"2f961c49c0f729340a6d885f8354c05af92b0a21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 22:17:38.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/77/check/gate-nova-docs-ubuntu-xenial/df98b66//doc/build/html/ : SUCCESS in 2m 30s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-nova-pep8-ubuntu-xenial/dc5aa15/ : SUCCESS in 6m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-nova-python27-db-ubuntu-xenial/0b2a7b7/ : SUCCESS in 10m 47s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/77/check/gate-nova-python35-db/5e2e386/ : SUCCESS in 7m 51s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0973720/ : FAILURE in 1h 12m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c900707/ : SUCCESS in 33m 05s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-nova-tox-db-functional-ubuntu-xenial/1d0578e/ : SUCCESS in 10m 13s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/ee59688/ : SUCCESS in 9m 43s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c247aeb/ : SUCCESS in 38m 41s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/77/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/44df4b2/ : FAILURE in 32m 42s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/d1100de/ : FAILURE in 57m 38s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-cells-ubuntu-xenial/a7a6821/ : SUCCESS in 37m 04s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/150fa12/ : SUCCESS in 1h 00m 56s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/ebaf2ff/ : FAILURE in 1h 04m 10s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/2c19a3a/ : SUCCESS in 55m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/77/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/db3e210/ : SUCCESS in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"768960df624c11acd1f252f77c6dd19a1ace2f83","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-24 22:20:40.000000000","message":"Patch Set 77:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/77/check/nova-pvm-dsvm-tempest-full/c2b7b5c : FAILURE in 1h 17m 19s","accounts_in_message":[],"_revision_number":77},{"id":"96c706cce8e4bc7b071d444af412d74affe50eb1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-24 22:27:23.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/77/check/check-dsvm-tempest-vz7-exe-minimal/61e3db6 : SUCCESS in 57m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":77},{"id":"143f4b9f914e228b03374b0d091cfefff69d1842","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-24 23:21:01.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/77/check/dsvm-tempest-neutron-network/2da3df5 : SUCCESS in 2h 03m 38s","accounts_in_message":[],"_revision_number":77},{"id":"fd98b27f168b974ec0774b2f139af760ef9347e3","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-24 23:28:45.000000000","message":"Patch Set 77:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/77/ : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":77},{"id":"19af823cbd76eda2987fe16bb527e97c1e9abe45","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 23:41:08.000000000","message":"Patch Set 77:\n\n(1 comment)","accounts_in_message":[],"_revision_number":77},{"id":"c52612fab521110532154ff5d043f5839f396f6f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-24 23:43:05.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"5d9ce1ae4dc0adb26b330bf1af8a239cbb0d9f40","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-25 00:05:02.000000000","message":"Patch Set 78:\n\n* pci-test http://52.27.155.124/pci/319379/78 : SUCCESS","accounts_in_message":[],"_revision_number":78},{"id":"4fc20cc1454943fbc8ce7c3a2c7cacda9f361bc1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 00:41:49.000000000","message":"Patch Set 78:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/78/check/check-dsvm-tempest-vz7-exe-minimal/4fba3d4 : SUCCESS in 57m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":78},{"id":"f268f451b141b4d8d0b26fd459e637f57bea8eab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-25 00:52:10.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/78/check/tempest-dsvm-full-xenial/703a86c/ : SUCCESS in 1h 07m 05s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/78/check/grenade-dsvm-xenial/5fabae2/ : SUCCESS in 38m 47s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"c663525607ef4c771e586dc855d2d3ba96e6c8f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 00:54:30.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/78/check/gate-nova-docs-ubuntu-xenial/8f5d497//doc/build/html/ : SUCCESS in 2m 23s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-nova-pep8-ubuntu-xenial/65cfd97/ : SUCCESS in 3m 58s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-nova-python27-db-ubuntu-xenial/962a00a/ : SUCCESS in 11m 51s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/78/check/gate-nova-python35-db/af39c02/ : SUCCESS in 9m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8320d2d/ : SUCCESS in 1h 07m 33s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4c6f878/ : SUCCESS in 34m 15s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-nova-tox-db-functional-ubuntu-xenial/2daf2c1/ : FAILURE in 10m 34s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/591b750/ : SUCCESS in 11m 02s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0492d38/ : SUCCESS in 36m 00s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/78/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/c60d4fa/ : FAILURE in 30m 33s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/be3cfaa/ : FAILURE in 50m 39s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-cells-ubuntu-xenial/ab4370c/ : SUCCESS in 37m 03s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ef8a596/ : SUCCESS in 1h 02m 35s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e0315c5/ : SUCCESS in 1h 08m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/9b1d61f/ : SUCCESS in 25m 38s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/78/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/56f7f23/ : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"0d339c6bf9852c2ff3a2d4c50968c083d7d79cfc","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-25 00:54:38.000000000","message":"Patch Set 78:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/78/ : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":78},{"id":"40f368126e6d060de5921bb7e483f0e3f1aea3dd","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-25 00:58:49.000000000","message":"Patch Set 78: Code-Review+2\n\nrecheck bug 1657087","accounts_in_message":[],"_revision_number":78},{"id":"8d8fbdc6f671e2172494b8bcca7809891d326175","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:01:09.000000000","message":"Patch Set 78:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/78/check/nova-pvm-dsvm-tempest-full/9132280 : FAILURE in 1h 16m 52s","accounts_in_message":[],"_revision_number":78},{"id":"890c8975caa96422ce94b4c301b3e29f11090f5f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 01:03:25.000000000","message":"Patch Set 78:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":78},{"id":"105071dcc11aa139ec9f5c8076f61ff128a0cfff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-25 01:03:28.000000000","message":"Patch Set 78: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":78},{"id":"4de8c9c8656771b504dd1fd10ec19c17255fe513","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-25 01:03:32.000000000","message":"Patch Set 78:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":78},{"id":"aaf3ce3f9bfd871f09b2b17b794cebd2b28cd66c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:03:40.000000000","message":"Patch Set 78:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial tempest-dsvm-full-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"a9fc0ebd32cbb8a54ec1143791195585d0f58a6c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:03:41.000000000","message":"Patch Set 78:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":78},{"id":"69852150e8542ee4a507d437821b97a57499961c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:03:46.000000000","message":"Patch Set 78:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":78},{"id":"f05e30c56f0ac2c62600f0f14846ec3c767a488c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:04:10.000000000","message":"Patch Set 75:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":75},{"id":"2363220037616988f8bdbe77b5310068ec36eb12","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-25 01:04:24.000000000","message":"Patch Set 77:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":77},{"id":"15f1fc2912c9238286f6c1ceb82a1a050dba75af","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-25 06:44:44.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/76/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/fd262d7 : SUCCESS in 46m 55s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":76},{"id":"857a20153c959d61021923f42b0f03db00aeffdb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-25 08:51:01.000000000","message":"Patch Set 76:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/76/check-nova/Nova-ML2-Sriov/4535e79 : SUCCESS in 43m 11s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/76/check-nova/Nova-MACVTAP-ML2-Sriov/4a2eee5 : SUCCESS in 48m 48s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":76},{"id":"107193e52b9b5d49cda96cad09483219e10f4e07","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 14:48:51.000000000","message":"Patch Set 78:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":78},{"id":"30324cae56a1cb527f91b9d41ac5fd672d4716af","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-01-25 15:08:11.000000000","message":"Patch Set 78:\n\n(1 comment)","accounts_in_message":[],"_revision_number":78},{"id":"c8a99e6091bb3215202d209a499f2e8ef1883cd7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 15:17:45.000000000","message":"Uploaded patch set 79.","accounts_in_message":[],"_revision_number":79},{"id":"bb5f34fcafb359735cb738429cea8469abb010d0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-25 15:57:46.000000000","message":"Patch Set 79:\n\n* pci-test http://52.27.155.124/pci/319379/79 : SUCCESS","accounts_in_message":[],"_revision_number":79},{"id":"a9a0216bd3b6e065ec9f0fa5ec4880d5c03a9495","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 16:21:27.000000000","message":"Patch Set 79:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/79/check/check-dsvm-tempest-vz7-exe-minimal/93f64b0 : FAILURE in 1h 02m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":79},{"id":"401b9e8230c5aec8c5af79e4c23d291653a9f97a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-25 16:26:17.000000000","message":"Patch Set 79:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/79/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/211c8fc : SUCCESS in 52m 18s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":79},{"id":"a71ea10d9a52ad1bc85d552dc20c4163c85996df","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-25 16:29:59.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/79/check/tempest-dsvm-full-xenial/d3e53c6/ : FAILURE in 1h 08m 58s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/79/check/grenade-dsvm-xenial/8870c6d/ : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"092d46d82e723bfdb07901e235831573761d811b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-25 16:43:29.000000000","message":"Patch Set 79:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/79/ : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":79},{"id":"dba04ebb2cc24fb7f9a124fb5ac88a6fed3d35e1","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-25 16:49:47.000000000","message":"Patch Set 79:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/79/check/nova-pvm-dsvm-tempest-full/f880c51 : FAILURE in 1h 30m 08s","accounts_in_message":[],"_revision_number":79},{"id":"00b09a5b5ebb921fb7991d7014d25b66c81d48bc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-25 17:07:25.000000000","message":"Patch Set 79:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/79/check-nova/Nova-ML2-Sriov/53abd59 : SUCCESS in 45m 27s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/79/check-nova/Nova-MACVTAP-ML2-Sriov/82861fa : SUCCESS in 44m 27s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":79},{"id":"8994abb38333a861dec2d169cbf7153fa3fcc050","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-25 17:10:30.000000000","message":"Patch Set 79:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/79/check/dsvm-tempest-neutron-network/1dd52f8 : SUCCESS in 1h 45m 59s","accounts_in_message":[],"_revision_number":79},{"id":"985d715db1ffc3900c539dd06c2731ca5c0ca533","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 17:22:29.000000000","message":"Patch Set 79:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial tempest-dsvm-full-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"5b4a08deba0a08e7e619d1d9aa061bc1ea3a3988","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 17:22:39.000000000","message":"Patch Set 79:\n\nThis change depends on a change that failed to merge.\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":79},{"id":"4edc3d286584907f654ce8ebee54c00499bf5709","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 17:22:44.000000000","message":"Patch Set 79: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":79},{"id":"8f05dbf836c38932e143ed12f7274a780e543489","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 18:20:06.000000000","message":"Patch Set 79:\n\nrecheck gdi jenkins, dep changed","accounts_in_message":[],"_revision_number":79},{"id":"4af25ccd374d6a325f045abeeb4cd58b22e70dc4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-25 19:31:30.000000000","message":"Patch Set 79:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/79/check/tempest-dsvm-full-xenial/d6eedd5/ : FAILURE in 1h 10m 45s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/79/check/grenade-dsvm-xenial/132082b/ : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"f9edf0cc5613d539d176e72dd6bf7d585744831b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-25 19:39:48.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"d9e52c2d46dda726833a7bfa9f4b01e86c911ab8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 20:18:25.000000000","message":"Patch Set 79:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/79/check/gate-nova-docs-ubuntu-xenial/161f7f0//doc/build/html/ : SUCCESS in 5m 56s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-nova-pep8-ubuntu-xenial/612c5ec/ : SUCCESS in 8m 55s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-nova-python27-db-ubuntu-xenial/03804ff/ : SUCCESS in 13m 39s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/79/check/gate-nova-python35-db/7c8b1f3/ : SUCCESS in 12m 16s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/b1ff370/ : FAILURE in 1h 02m 14s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-grenade-dsvm-neutron-ubuntu-xenial/cdd9b0e/ : SUCCESS in 37m 04s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-nova-tox-db-functional-ubuntu-xenial/159517a/ : SUCCESS in 18m 08s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/2778483/ : SUCCESS in 22m 16s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/dd116d4/ : SUCCESS in 39m 12s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/79/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/73d1d0c/ : FAILURE in 52m 16s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/f3b0878/ : FAILURE in 1h 13m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-cells-ubuntu-xenial/62d7333/ : SUCCESS in 50m 33s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/8fc229e/ : FAILURE in 1h 28m 58s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/2530f02/ : SUCCESS in 1h 05m 05s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/935b01b/ : SUCCESS in 29m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/79/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8695c80/ : SUCCESS in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"1875558c0f6e07507a03e2fccd1f88719703cc55","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-25 20:49:49.000000000","message":"Patch Set 79:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/79/ : FAILURE in 36m 28s","accounts_in_message":[],"_revision_number":79},{"id":"5eda26605543263b23496cf495c53aff70ca9f90","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 21:13:11.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"39e8d6a1d69d9ed15223dccf08532631712d9424","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 21:39:13.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"7ca89402b5c30a0834c6374662b52e4293dfde45","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-25 22:07:02.000000000","message":"Patch Set 80:\n\n* pci-test http://52.27.155.124/pci/319379/80 : SUCCESS","accounts_in_message":[],"_revision_number":80},{"id":"e4eedbff55362d93a8717059eea96821c7dcbf23","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 22:31:53.000000000","message":"Patch Set 79:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":79},{"id":"2b1ad37ce3a88bfff1f46670013b9bcb2012543e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 22:36:31.000000000","message":"Patch Set 80:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/80/check/check-dsvm-tempest-vz7-exe-minimal/53ff246 : SUCCESS in 55m 55s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":80},{"id":"f0ab5cdc47b6a08facb70695913e0c0548f90272","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-25 22:44:29.000000000","message":"Patch Set 80:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/80/check/nova-pvm-dsvm-tempest-full/4ceb9fa : FAILURE in 1h 03m 55s","accounts_in_message":[],"_revision_number":80},{"id":"727c6f2e2ddc786c6abc1b40c2546263931c3c25","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-25 22:54:15.000000000","message":"Patch Set 80: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/80/check/tempest-dsvm-full-xenial/a8a30fa/ : FAILURE in 1h 13m 00s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/80/check/grenade-dsvm-xenial/6096c2e/ : SUCCESS in 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"69b078807cab23ea0001e417ccdbad7cda93e278","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 23:11:04.000000000","message":"Patch Set 80: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/80/check/gate-nova-docs-ubuntu-xenial/96d51d5//doc/build/html/ : SUCCESS in 3m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-nova-pep8-ubuntu-xenial/b1735d9/ : SUCCESS in 5m 48s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-nova-python27-db-ubuntu-xenial/2a91ff4/ : SUCCESS in 7m 04s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/80/check/gate-nova-python35-db/29a3049/ : SUCCESS in 10m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e5d22e9/ : FAILURE in 1h 28m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b49449f/ : SUCCESS in 32m 41s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-nova-tox-db-functional-ubuntu-xenial/034f155/ : SUCCESS in 10m 22s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/70ed163/ : SUCCESS in 10m 25s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/dbc334a/ : SUCCESS in 37m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/80/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/487d272/ : FAILURE in 32m 54s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/3a87f7d/ : FAILURE in 1h 11m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-cells-ubuntu-xenial/b8627c6/ : SUCCESS in 34m 23s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4d0ba7a/ : SUCCESS in 1h 26m 58s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/68d4d5b/ : SUCCESS in 1h 13m 21s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/1ca8a17/ : SUCCESS in 24m 35s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/80/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/11b6292/ : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"0b8ab70fbdc5548371b21ba676f17a319f8cec70","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-25 23:19:44.000000000","message":"Patch Set 80:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/80/ : FAILURE in 1h 39m 07s","accounts_in_message":[],"_revision_number":80},{"id":"e78769757b4a041834815a0d14d70def8c7fcbb4","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-25 23:35:51.000000000","message":"Patch Set 80:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/80/check/dsvm-tempest-kvm/d9ca198 : SUCCESS in 1h 03m 01s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/80/check/dsvm-tempest-exe/daa6678 : SUCCESS in 45m 22s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":80},{"id":"b54c20e5c9626361816eae6e289d677bb0d6d91c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-25 23:36:21.000000000","message":"Uploaded patch set 81.","accounts_in_message":[],"_revision_number":81},{"id":"3081de81b6e8974e13863139422e2d62404733c5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-26 00:16:21.000000000","message":"Patch Set 81:\n\n* pci-test http://52.27.155.124/pci/319379/81 : SUCCESS","accounts_in_message":[],"_revision_number":81},{"id":"8f92633b084ad809b972c048fcadadd4ed432d8d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 00:34:36.000000000","message":"Patch Set 81:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/81/check/check-dsvm-tempest-vz7-exe-minimal/e485ae7 : SUCCESS in 57m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":81},{"id":"33ddb08058c702a4010cb8d36bd79cc3134eff52","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 00:36:05.000000000","message":"Patch Set 81:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/81/check/dsvm-tempest-kvm/38aab53 : FAILURE in 58m 02s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/81/check/dsvm-tempest-exe/0ff0513 : SUCCESS in 45m 20s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":81},{"id":"47934626ec33dc9cd0c6093c48d38d4b898c49e4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 00:41:42.000000000","message":"Patch Set 81: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/81/check/tempest-dsvm-full-xenial/d3ab57a/ : FAILURE in 1h 04m 28s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/81/check/grenade-dsvm-xenial/c042c05/ : SUCCESS in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"e7b16682e2962212fc1cb8924ed5803ac06f35a3","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 00:59:41.000000000","message":"Patch Set 81:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/81/ : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":81},{"id":"b35a2589271a51a99de90af5ea52cb996a80643b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-26 01:03:05.000000000","message":"Patch Set 81:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/81/check/nova-pvm-dsvm-tempest-full/9c2e163 : FAILURE in 1h 21m 31s","accounts_in_message":[],"_revision_number":81},{"id":"a5ec2c065e357ff0e5e8d1f3caec8138ca1ac46c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 01:04:12.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/81/check/gate-nova-docs-ubuntu-xenial/eef59ff//doc/build/html/ : SUCCESS in 3m 20s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-nova-pep8-ubuntu-xenial/b48db64/ : SUCCESS in 4m 03s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-nova-python27-db-ubuntu-xenial/10f0b53/ : SUCCESS in 7m 21s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/81/check/gate-nova-python35-db/27de1ac/ : SUCCESS in 11m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0f0d4e1/ : SUCCESS in 59m 06s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-grenade-dsvm-neutron-ubuntu-xenial/5d35b41/ : SUCCESS in 34m 41s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-nova-tox-db-functional-ubuntu-xenial/369214a/ : SUCCESS in 11m 04s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/bbd09ae/ : SUCCESS in 15m 45s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e7da1f5/ : SUCCESS in 38m 11s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/81/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/cc5618d/ : FAILURE in 31m 11s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/e82d299/ : FAILURE in 57m 53s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-cells-ubuntu-xenial/83aafed/ : SUCCESS in 36m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/42f4d79/ : SUCCESS in 1h 24m 08s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/5dd5a01/ : SUCCESS in 1h 08m 45s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/7797d8f/ : SUCCESS in 24m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/81/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/cd9c3d4/ : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"b78bbfa9db57f2435d46aa3ea69036b10a2a9696","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 01:20:46.000000000","message":"Patch Set 81:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-25/319379/81/check/tempest-dsvm-full-nfv-xenial/eb551fc : SUCCESS in 1h 42m 56s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-25/319379/81/check/tempest-dsvm-intel-nfv-xenial/f68f1e8 : SUCCESS in 42m 11s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-25/319379/81/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/14d9b31 : FAILURE in 1h 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"894828c75e9193c002c3f0298103165ff739f5c2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-26 01:26:51.000000000","message":"Patch Set 81:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/81/check/dsvm-tempest-neutron-network/40b7d44 : FAILURE in 1h 48m 57s","accounts_in_message":[],"_revision_number":81},{"id":"c5923a7219cd84bbcf187129ad7ae93fa6cf628c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 01:55:10.000000000","message":"Patch Set 81: Code-Review+2","accounts_in_message":[],"_revision_number":81},{"id":"731e804a094db74fefcf8c9ac0ec3b4304bb85c9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 02:03:21.000000000","message":"Patch Set 81: -Code-Review\n\nHm, it looks like the IBM PowerKVM CI failure may be legit: \"OrphanedObjectError: Cannot call destroy on orphaned Instance object\" raised from L1790 where instance.destroy() is called on not the looked up _instance.\n\nhttp://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/81/check/tempest-dsvm-full-xenial/d3ab57a/screen-n-api.txt.gz","accounts_in_message":[],"_revision_number":81},{"id":"92c483cfb3dbf79d8a6f7564cf04c2f36813f795","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 02:06:59.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"a1ad8ac2c643bfa0e657877fd9c670168e08bc4a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 02:21:45.000000000","message":"Patch Set 81:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/81/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/410a383 : SUCCESS in 55m 49s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":81},{"id":"c880eb5378a125490a0d91a1c7ac7f5fd393d5f2","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 02:26:31.000000000","message":"Patch Set 81: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":81},{"id":"63a0c8771437499a90d4cf7b3d38af0e5ec3596c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 02:26:40.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"7f04566a74cdf6efff1b17e6202694cffcddf0ba","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 02:33:48.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"daace1a520cf886dc8a7f14f40fb58423c5fc27a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 02:38:21.000000000","message":"Patch Set 81:\n\n(2 comments)","accounts_in_message":[],"_revision_number":81},{"id":"9d664ef138b9e3b6aca1aee926e6f5518157055e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 02:40:17.000000000","message":"Patch Set 81:\n\n(1 comment)","accounts_in_message":[],"_revision_number":81},{"id":"2f8e3faa05e74c62b9b5450b2ac595b9d4e2e1ed","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 02:58:48.000000000","message":"Patch Set 81:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/81/check-nova/Nova-ML2-Sriov/d91a8f9 : SUCCESS in 42m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/81/check-nova/Nova-MACVTAP-ML2-Sriov/bb8f1f4 : SUCCESS in 47m 06s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":81},{"id":"90c31c2192c91456a0fa1d365d03674b5f625f25","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 03:02:08.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"7df11c0f55008174800c6574f2c208c43ce67013","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-26 03:32:12.000000000","message":"Patch Set 82:\n\n* pci-test http://52.27.155.124/pci/319379/82 : SUCCESS","accounts_in_message":[],"_revision_number":82},{"id":"04b69fb9f11e4ba2f613a61e98d687dcfd8967c5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 03:58:51.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/82/check/check-dsvm-tempest-vz7-exe-minimal/6411484 : SUCCESS in 55m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":82},{"id":"3e28e687b982d13a5cac1329e34253d659dfe8f9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 04:01:03.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/82/check/tempest-dsvm-full-xenial/04627cd/ : SUCCESS in 58m 02s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/82/check/grenade-dsvm-xenial/5239810/ : SUCCESS in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"cb8917601de46b43e77ff3fc350a09b634fcf207","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 04:01:32.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/82/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/1f6c65e : SUCCESS in 51m 35s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":82},{"id":"41178a3cc11b8ea991d4d76c11774af7b9191850","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 04:01:47.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/82/check/dsvm-tempest-kvm/909675e : SUCCESS in 58m 04s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/82/check/dsvm-tempest-exe/9513418 : SUCCESS in 44m 25s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":82},{"id":"06a54509b5016247b6dd17b1f9a1bb1d813ae22e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-26 04:17:32.000000000","message":"Patch Set 82:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/82/check/nova-pvm-dsvm-tempest-full/a69f297 : FAILURE in 1h 09m 47s","accounts_in_message":[],"_revision_number":82},{"id":"fc00dcef4bb36e10bef5592ed81ed11a8ec0ec14","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 04:25:44.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/82/check-nova/Nova-ML2-Sriov/e6e2258 : SUCCESS in 45m 59s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/82/check-nova/Nova-MACVTAP-ML2-Sriov/8cd4ecf : SUCCESS in 42m 36s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":82},{"id":"811ab6e14ef4a012298966716bcfd76ac9892dbf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 04:28:47.000000000","message":"Patch Set 82: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/82/check/gate-nova-docs-ubuntu-xenial/4408213//doc/build/html/ : SUCCESS in 2m 20s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-pep8-ubuntu-xenial/471f3b8/ : SUCCESS in 5m 50s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-python27-db-ubuntu-xenial/bc75c03/ : SUCCESS in 10m 57s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/82/check/gate-nova-python35-db/16c7ff2/ : SUCCESS in 7m 16s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/72d1a95/ : FAILURE in 1h 06m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a8c5bcd/ : SUCCESS in 50m 43s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-tox-db-functional-ubuntu-xenial/90c26e1/ : SUCCESS in 19m 52s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/f46d9a5/ : SUCCESS in 15m 51s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/92bb503/ : SUCCESS in 1h 11m 05s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/ed312e1/ : FAILURE in 32m 14s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/a9e840d/ : SUCCESS in 1h 22m 56s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-cells-ubuntu-xenial/214f377/ : SUCCESS in 53m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/eec3121/ : FAILURE in 1h 21m 27s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3001771/ : SUCCESS in 1h 08m 11s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d8c7338/ : SUCCESS in 23m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4d7df97/ : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"eb495d6a8b2d910e9e21c67dbc97eb44bef16894","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 04:29:29.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/82/ : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":82},{"id":"b1f88dcdf5eb8ad6e3d412e8f78a27c76c1285e3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 04:31:52.000000000","message":"Patch Set 82:\n\nrecheck weird sql fail","accounts_in_message":[],"_revision_number":82},{"id":"d7ea58a1110a01e63b924d93269b0a905a96cbb0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-26 04:38:49.000000000","message":"Patch Set 81:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-81 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":81},{"id":"ee95807502e2905c4d43bf9d51f8ae75806c80ee","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 04:44:15.000000000","message":"Patch Set 82:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/82/check/tempest-dsvm-full-nfv-xenial/af50ef3 : SUCCESS in 1h 40m 28s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/82/check/tempest-dsvm-intel-nfv-xenial/62a0d91 : SUCCESS in 43m 12s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/82/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/1b8e6e7 : FAILURE in 1h 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"03ed0fd33af5a4993f56591cadeecae1cae376de","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-26 04:54:42.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/82/check/dsvm-tempest-neutron-network/65b1842 : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":82},{"id":"8e63b95b917e0629857a35eaf5c1351750d2083c","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 05:45:52.000000000","message":"Patch Set 82:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/82/ : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":82},{"id":"2bc39424662cbcf481e58d8258938d98a63789bb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 05:54:20.000000000","message":"Patch Set 82:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/82/check/tempest-dsvm-full-xenial/a11510d/ : SUCCESS in 1h 21m 31s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/82/check/grenade-dsvm-xenial/bef2c9c/ : SUCCESS in 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"771e1e393e7f04f39cc35d8f04d656d4da8efce7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 06:10:34.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/82/check/gate-nova-docs-ubuntu-xenial/d97be2f//doc/build/html/ : SUCCESS in 2m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-pep8-ubuntu-xenial/12e0b5c/ : SUCCESS in 3m 53s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-python27-db-ubuntu-xenial/3e41356/ : SUCCESS in 6m 48s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/82/check/gate-nova-python35-db/bcabf9e/ : SUCCESS in 6m 43s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9637ae2/ : SUCCESS in 1h 17m 24s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1d84945/ : SUCCESS in 53m 14s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-tox-db-functional-ubuntu-xenial/5d512de/ : SUCCESS in 17m 52s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/1833b4d/ : SUCCESS in 14m 48s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ec2da09/ : SUCCESS in 35m 18s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/82/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/86678d3/ : FAILURE in 31m 25s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/d02aec0/ : FAILURE in 48m 28s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-cells-ubuntu-xenial/768c54b/ : SUCCESS in 52m 53s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0b25d93/ : SUCCESS in 57m 27s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/5dcb106/ : SUCCESS in 1h 36m 11s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8c6d287/ : SUCCESS in 24m 06s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/1af0ce3/ : SUCCESS in 30m 19s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"029bce2707e8bf309806ae22c39783d5aa22cc9f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 06:33:35.000000000","message":"Patch Set 82:\n\nI noticed some traces in n-cond while destroying an instance while it was building (\"host changed\" during destroy):\n\nhttp://logs.openstack.org/79/319379/82/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9637ae2/logs/screen-n-cond.txt.gz?#_2017-01-26_05_11_04_047","accounts_in_message":[],"_revision_number":82},{"id":"4f900be49c17e7865b9e099d39eda2457cc00650","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 06:52:42.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"fc9be025ba75084f99a6205bafaf5b025e446db7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-26 07:21:53.000000000","message":"Patch Set 83:\n\n* pci-test http://52.27.155.124/pci/319379/83 : SUCCESS","accounts_in_message":[],"_revision_number":83},{"id":"83704a6b447cb5370c88e96b7e4700ecf05a325d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-26 07:29:38.000000000","message":"Patch Set 82:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-82 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":82},{"id":"28a4a8983bbb9fa6a6115ccd1fc315035f591ee5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 07:50:28.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/83/check/dsvm-tempest-kvm/e639623 : SUCCESS in 56m 51s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/79/319379/83/check/dsvm-tempest-exe/d40ba2c : SUCCESS in 41m 39s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":83},{"id":"f8a152d0fd57f6117bfb006bb5b6c4945b407e56","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 07:55:08.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/83/check/check-dsvm-tempest-vz7-exe-minimal/c59f827 : SUCCESS in 1h 01m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":83},{"id":"693ffd3c0444e6a3a885170868289ce7ce1612dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 08:03:55.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/83/check/tempest-dsvm-full-xenial/b4b452e/ : SUCCESS in 1h 08m 28s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/83/check/grenade-dsvm-xenial/ab0ebd2/ : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"9c64a648fa8a4b6f345bca0d1365d04f3a57fd7b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 08:09:04.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/83/ : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":83},{"id":"265b8d5e18cd7dd2e971e14aee8c1fcc963b70ac","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 08:10:13.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/83/check-nova/Nova-ML2-Sriov/3b1bffa : SUCCESS in 54m 22s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/83/check-nova/Nova-MACVTAP-ML2-Sriov/7109397 : SUCCESS in 58m 03s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":83},{"id":"3349a8d51d0361052029825b78d4bcff57b26713","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 08:21:39.000000000","message":"Patch Set 83: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/83/check/gate-nova-docs-ubuntu-xenial/f05e378//doc/build/html/ : SUCCESS in 4m 19s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-pep8-ubuntu-xenial/96e0c97/ : SUCCESS in 3m 39s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-python27-db-ubuntu-xenial/e10717b/ : SUCCESS in 7m 04s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/83/check/gate-nova-python35-db/c053708/ : SUCCESS in 8m 02s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e3c51ab/ : SUCCESS in 1h 01m 02s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-ubuntu-xenial/df90511/ : SUCCESS in 31m 07s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-tox-db-functional-ubuntu-xenial/42bce23/ : FAILURE in 16m 48s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/6f9c458/ : SUCCESS in 10m 07s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e4b4712/ : SUCCESS in 36m 28s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/aeee7f9/ : FAILURE in 32m 11s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/70daa4f/ : SUCCESS in 58m 57s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-cells-ubuntu-xenial/c7c1967/ : SUCCESS in 57m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/c4df77b/ : FAILURE in 1h 27m 19s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/d174a99/ : SUCCESS in 1h 14m 21s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/62a0a05/ : SUCCESS in 23m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/427b99c/ : SUCCESS in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"711506370a2460cfbec921dbe212eebd4fd3eba7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 08:21:45.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/83/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/8443922 : FAILURE in 1h 17m 59s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":83},{"id":"3f4aa7b06a77bd4fa4d0d6f5f05917af78d224d8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-26 08:25:39.000000000","message":"Patch Set 83:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/83/check/nova-pvm-dsvm-tempest-full/b037f7f : FAILURE in 1h 27m 27s","accounts_in_message":[],"_revision_number":83},{"id":"064eac7fa32b0336170d03aa84c4e7a180f650ef","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 08:27:09.000000000","message":"Patch Set 83:\n\nrecheck bug 1657087","accounts_in_message":[],"_revision_number":83},{"id":"e5ce3d7f0c21aefcfb1298d4df5881ee1302aa71","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 08:34:39.000000000","message":"Patch Set 83:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/83/check/tempest-dsvm-full-nfv-xenial/7f93676 : SUCCESS in 1h 39m 53s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/83/check/tempest-dsvm-intel-nfv-xenial/7cfe432 : SUCCESS in 44m 03s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/83/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/ecf9ede : FAILURE in 1h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"7e22c8643de61f59b58149d8b4359bd56ad57243","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-26 08:36:44.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/83/check/dsvm-tempest-neutron-network/62ac5f8 : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":83},{"id":"36f4de112dba860167f567ceff5f24e61a1ea0b0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 09:40:53.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/83/check-nova/Nova-ML2-Sriov/c3aa040 : SUCCESS in 57m 35s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/83/check-nova/Nova-MACVTAP-ML2-Sriov/232e61c : SUCCESS in 59m 47s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":83},{"id":"0e935a1594ecf3acf2236b5bf11bb9ee9fa30a99","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 09:47:49.000000000","message":"Patch Set 83:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/83/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/469546f : SUCCESS in 1h 09m 10s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":83},{"id":"9d9fa4b20821d29894ca1133b061475c43cca5fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 09:48:07.000000000","message":"Patch Set 83:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/83/check/tempest-dsvm-full-xenial/392243c/ : SUCCESS in 1h 20m 30s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/83/check/grenade-dsvm-xenial/30532a3/ : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"df49ea71b44ccd7065ef3f99b0cc0b58e6b180a4","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 10:08:44.000000000","message":"Patch Set 83:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/83/ : FAILURE in 1h 40m 45s","accounts_in_message":[],"_revision_number":83},{"id":"fa06b9559d5d64d4035d8dfa6887821319994b82","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 10:15:55.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/83/check/gate-nova-docs-ubuntu-xenial/113bf65//doc/build/html/ : SUCCESS in 2m 29s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-pep8-ubuntu-xenial/31afdf4/ : SUCCESS in 3m 32s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-python27-db-ubuntu-xenial/78a6cdf/ : SUCCESS in 10m 47s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/83/check/gate-nova-python35-db/3c84d92/ : SUCCESS in 10m 41s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e950861/ : SUCCESS in 56m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e891aa1/ : SUCCESS in 36m 00s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-tox-db-functional-ubuntu-xenial/394859f/ : SUCCESS in 16m 58s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/1c80803/ : SUCCESS in 13m 54s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/4115866/ : SUCCESS in 37m 02s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/83/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/1738de8/ : FAILURE in 59m 45s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/8d24078/ : SUCCESS in 56m 35s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-cells-ubuntu-xenial/4fd754f/ : SUCCESS in 35m 56s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b809d9d/ : SUCCESS in 1h 02m 41s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/77e99b9/ : SUCCESS in 1h 45m 40s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/87b0dcd/ : SUCCESS in 24m 40s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/83/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4e46ae9/ : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"4a3a1d4d8a8cb8c930bb46522375689444f87b39","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-26 11:07:54.000000000","message":"Patch Set 83:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-83 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":83},{"id":"3042998dd005774ad1e20129137c558b116a0f73","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-01-26 11:20:56.000000000","message":"Patch Set 83: Code-Review+2\n\n(3 comments)\n\nA few comments but nothing blocking for us merging it.","accounts_in_message":[],"_revision_number":83},{"id":"5b40e7e74fe088d461bacca5bf5d0a4c98ce3a77","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-01-26 14:35:52.000000000","message":"Patch Set 83: Code-Review+2 Workflow+1\n\n(1 comment)\n\nSo this makes sense to my reading of the situation. Lets get this mighty thing landed now.","accounts_in_message":[],"_revision_number":83},{"id":"9e054059d1a10e5c2576cbfec983cccd1cd598a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 14:36:06.000000000","message":"Patch Set 83: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":83},{"id":"edd808f8f849c7094ed9b128dcb9387e953b77bb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2017-01-26 16:59:45.000000000","message":"Patch Set 83:\n\n(1 comment)","accounts_in_message":[],"_revision_number":83},{"id":"533f6dc9bacef6e8941e3cb9776fca6e9d70226c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 19:56:14.000000000","message":"Uploaded patch set 84: Patch Set 83 was rebased.","accounts_in_message":[],"_revision_number":84},{"id":"d98712942747ad2db25ec0a52cfbdcf981a6ade3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-01-26 19:57:02.000000000","message":"Patch Set 84: Code-Review+2 Workflow+1\n\nPreviously approved, hit bug 1657087 in the gate, so re-queuing.","accounts_in_message":[],"_revision_number":84},{"id":"6807bb5891f9f021d07ac33dc968afa60f11335e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-26 20:17:32.000000000","message":"Patch Set 84:\n\n* pci-test http://52.27.155.124/pci/319379/84 : SUCCESS","accounts_in_message":[],"_revision_number":84},{"id":"0e9b71393e15e214e391136101ce3184e798b073","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 21:05:36.000000000","message":"Patch Set 84:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/79/319379/84/check/check-dsvm-tempest-vz7-exe-minimal/b10d65c : SUCCESS in 59m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":84},{"id":"271926a2feda53bb0db7221a414401e835514989","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-26 21:08:20.000000000","message":"Patch Set 84: Verified-1\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/84/check/tempest-dsvm-full-xenial/5cdef02/ : FAILURE in 1h 11m 27s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/79/319379/84/check/grenade-dsvm-xenial/3ca446e/ : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"c385796f797f015e1211069a7b7dff682f520844","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-26 21:18:20.000000000","message":"Patch Set 84:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/319379/84/ : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":84},{"id":"40abc66cca595376064f39b5a8b50c984b9df96b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-26 21:18:48.000000000","message":"Patch Set 84:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-pvm-dsvm-tempest-full http://184.172.12.213/79/319379/84/check/nova-pvm-dsvm-tempest-full/32e83dd : FAILURE in 1h 17m 36s","accounts_in_message":[],"_revision_number":84},{"id":"34837fda0410bbb4bca703fc57a3e308814c3523","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-01-26 21:28:57.000000000","message":"Patch Set 84:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/319379/84/check/dsvm-tempest-neutron-network/3a14c92 : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":84},{"id":"5cd2357dc4ee4cea2fdf312e859a9ddcd80c2895","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 21:29:04.000000000","message":"Patch Set 84: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/84/check/gate-nova-docs-ubuntu-xenial/af893ce//doc/build/html/ : SUCCESS in 5m 03s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-nova-pep8-ubuntu-xenial/4ac8fee/ : SUCCESS in 6m 10s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-nova-python27-db-ubuntu-xenial/f3a15a0/ : SUCCESS in 11m 14s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/84/check/gate-nova-python35-db/d023795/ : SUCCESS in 9m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/9a282b2/ : SUCCESS in 1h 29m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3cc37ea/ : SUCCESS in 49m 27s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-nova-tox-db-functional-ubuntu-xenial/a0aa23c/ : SUCCESS in 11m 04s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/ce5187a/ : SUCCESS in 10m 35s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0d8a609/ : SUCCESS in 1h 00m 20s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/79/319379/84/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/9cce811/ : FAILURE in 30m 36s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/f2fdec8/ : SUCCESS in 1h 18m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-cells-ubuntu-xenial/d134880/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/f790555/ : SUCCESS in 1h 21m 49s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/5b3241f/ : SUCCESS in 1h 02m 16s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/62b6ba0/ : SUCCESS in 24m 59s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/319379/84/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4ff2855/ : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"be2a213fea083cad649c8cc997c28bea7af36737","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 21:29:16.000000000","message":"Patch Set 84: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":84},{"id":"3c5d86001a1923a9b434b322f640d015340cc492","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-26 21:47:32.000000000","message":"Patch Set 84:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/84/check/tempest-dsvm-full-nfv-xenial/61d2699 : SUCCESS in 1h 47m 54s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/84/check/tempest-dsvm-intel-nfv-xenial/8c9bcca : SUCCESS in 43m 14s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-26/319379/84/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/966ed94 : FAILURE in 1h 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"ab232b6fd1d43e14a9b8ac4361b3c27d245bcf3d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-26 22:59:27.000000000","message":"Patch Set 84:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/79/319379/84/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/ac320ad : FAILURE in 47m 55s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":84},{"id":"b984875479ed4c315fed6a2dcc2c535575c345fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-27 01:09:29.000000000","message":"Patch Set 84:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-79-319379-84 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":84},{"id":"b79b6c15f4c37db7d748c96cf8010d72c457d58f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-27 01:50:06.000000000","message":"Patch Set 84:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/79/319379/84/check-nova/Nova-ML2-Sriov/c52e8b4 : SUCCESS in 43m 24s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/79/319379/84/check-nova/Nova-MACVTAP-ML2-Sriov/248d606 : SUCCESS in 46m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":84},{"id":"a58e11034bcf288c19f426bd6f1110af4aca5c33","date":"2017-01-27 08:47:06.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":84},{"id":"20dfb05770370ac3b405252539824adead37aae9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-27 08:47:06.000000000","message":"Patch Set 84: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/79/319379/84/gate/gate-nova-docs-ubuntu-xenial/a8b5697//doc/build/html/ : SUCCESS in 2m 24s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-nova-pep8-ubuntu-xenial/de91629/ : SUCCESS in 3m 41s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-nova-python27-db-ubuntu-xenial/95d111f/ : SUCCESS in 7m 27s\n- gate-nova-python35-db http://logs.openstack.org/79/319379/84/gate/gate-nova-python35-db/3ce6cf9/ : SUCCESS in 6m 52s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/37a3aee/ : SUCCESS in 1h 18m 36s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/5b3abcb/ : SUCCESS in 30m 36s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-nova-tox-db-functional-ubuntu-xenial/f9f3e6c/ : SUCCESS in 10m 17s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-tempest-dsvm-cells-ubuntu-xenial/c6d8ff5/ : SUCCESS in 36m 35s\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/79/319379/84/gate/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/75a26d9/ : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":84},{"id":"29d8dd90a3ea8764592d83df5a6fa4de72e7c73e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-01-29 00:39:18.000000000","message":"Patch Set 84: Reverted\n\nThis patchset was reverted in change: Iac615e03cdee8a3a6abffbe33b1e77a1f73e4281","accounts_in_message":[],"_revision_number":84},{"id":"c47a1fedd5ecf5fd2a9aa5764aa5aa982fe8bf80","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-31 11:09:04.000000000","message":"Patch Set 54:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/54/2520/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":54},{"id":"5466d2a28d5f02717c6e04a8155a52473c50455f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-01 10:22:59.000000000","message":"Patch Set 55:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/55/2545/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":55},{"id":"ec20228d86fce234e6c52069abaa524853fd4423","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-04 18:39:01.000000000","message":"Patch Set 57:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/57/2677/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":57},{"id":"e2ed5a8ba9a40de454a69fc346a2da968c765188","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-05 21:45:38.000000000","message":"Patch Set 58:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/58/2726/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":58},{"id":"592ad45da1cbd7476afe6fced90af5497b0ea7d2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-08 16:22:15.000000000","message":"Patch Set 72:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/72/2847/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":72},{"id":"fba0f244d5a5120c5ea7bd28f866f698b02ac266","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-09 19:53:47.000000000","message":"Patch Set 59:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/59/2895/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":59},{"id":"c7536d6f6ebbe2548b69f1b5faac194fd1af048d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-09 23:34:39.000000000","message":"Patch Set 60:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/60/2902/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":60},{"id":"a67a3d645082e757727ae6289249d014a92ac57e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-09 23:43:26.000000000","message":"Patch Set 61:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/61/2903/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":61},{"id":"85ce3d639c4d499a5e7b85f080ec717f56609672","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-10 02:54:39.000000000","message":"Patch Set 62:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/62/2909/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":62},{"id":"5c75c04886493336473678a576f4e9dbafd06ee4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-10 14:35:00.000000000","message":"Patch Set 63:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/63/2923/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":63},{"id":"31bc6dab3daa972e16cecb1493d8637ef620dc02","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-10 16:13:00.000000000","message":"Patch Set 63:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/63/2925/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":63},{"id":"a42ca0a640950faf7cd666267a4e58e3cc6c8e46","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-26 12:55:08.000000000","message":"Patch Set 64:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/64/3376/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":64},{"id":"ac84656a66e0aa3e2eaa99676d40e6610d3b76eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-02-26 18:06:26.000000000","message":"Patch Set 70:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/70/3393/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":70},{"id":"cc2f3b54c22a7b802b3f3751a7273f43913b2693","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-03-04 12:57:35.000000000","message":"Patch Set 66:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/66/3549/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":66},{"id":"cf70533266d7fbc76c3f4b9bfc9a88e321b5d35c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-03-04 13:44:45.000000000","message":"Patch Set 67:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/67/3556/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":67},{"id":"bbe21dd921564ceaedaf05856209cb34a81eca6f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-03-16 01:00:48.000000000","message":"Patch Set 68:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/68/3686/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":68},{"id":"43c49d631f8e427b2a74aae14666c3cc58b127e9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-03-16 03:13:02.000000000","message":"Patch Set 69:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/69/3695/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":69},{"id":"6511dd200e8e44a83be77320393798f416e7fa51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-03-18 01:37:30.000000000","message":"Patch Set 71:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/319379/71/3785/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":71}],"current_revision_number":84,"current_revision":"3c2ce9450bea8d417bfd3fbe26ee688446c6a348","revisions":{"3b6440b015c87e7bfcb82f57c4cf9e8f25b2abd2":{"kind":"REWORK","_number":1,"created":"2016-05-20 17:17:50.000000000","uploader":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"ref":"refs/changes/79/319379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/1"}}},"commit":{"parents":[{"commit":"446f93d7c7a4fe0809b0cb4e0091d54a1b959974","subject":"Create Instance from BuildRequest if not in a cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/446f93d7c7a4fe0809b0cb4e0091d54a1b959974"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-05-20 17:17:23.000000000","tz":-240},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nPartially-implements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b6440b015c87e7bfcb82f57c4cf9e8f25b2abd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b6440b015c87e7bfcb82f57c4cf9e8f25b2abd2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"446f93d7c7a4fe0809b0cb4e0091d54a1b959974","is_merged_in_target_branch":false,"change_id":"Ib984c30543acb3ca9cb95fb53d44d9ded0f5a5c8","change_number":263927,"patch_set_number":32,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c754898b1e214e34688fd76700b1f49907c5c301":{"kind":"REWORK","_number":2,"created":"2016-08-30 19:00:47.000000000","uploader":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"ref":"refs/changes/79/319379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/2"}}},"commit":{"parents":[{"commit":"c1fc1a2b12a59d9f034fd328c8f82281dab08f35","subject":"Pull from cell0 and build_requests for instance list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1fc1a2b12a59d9f034fd328c8f82281dab08f35"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-08-30 18:41:26.000000000","tz":-240},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c754898b1e214e34688fd76700b1f49907c5c301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c754898b1e214e34688fd76700b1f49907c5c301"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1fc1a2b12a59d9f034fd328c8f82281dab08f35","is_merged_in_target_branch":false,"change_id":"I625c7714fb04eb2bf07695fd7ec053a2f32c62bc","change_number":356138,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"796faa011a54d9bc165c87b651003b3411a9b974":{"kind":"REWORK","_number":3,"created":"2016-08-30 20:49:14.000000000","uploader":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"ref":"refs/changes/79/319379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/3"}}},"commit":{"parents":[{"commit":"c1fc1a2b12a59d9f034fd328c8f82281dab08f35","subject":"Pull from cell0 and build_requests for instance list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1fc1a2b12a59d9f034fd328c8f82281dab08f35"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-08-30 20:01:54.000000000","tz":-240},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/796faa011a54d9bc165c87b651003b3411a9b974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/796faa011a54d9bc165c87b651003b3411a9b974"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1fc1a2b12a59d9f034fd328c8f82281dab08f35","is_merged_in_target_branch":false,"change_id":"I625c7714fb04eb2bf07695fd7ec053a2f32c62bc","change_number":356138,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9048088ab529c95d1b964f0d9bb8ad4900553323":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-09-01 15:03:49.000000000","uploader":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"ref":"refs/changes/79/319379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/4"}}},"commit":{"parents":[{"commit":"184a703e0671ebb18121d439b8b87dd5c0e696d8","subject":"Pull from cell0 and build_requests for instance list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/184a703e0671ebb18121d439b8b87dd5c0e696d8"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-09-01 15:03:00.000000000","tz":-240},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9048088ab529c95d1b964f0d9bb8ad4900553323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9048088ab529c95d1b964f0d9bb8ad4900553323"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"184a703e0671ebb18121d439b8b87dd5c0e696d8","is_merged_in_target_branch":false,"change_id":"I625c7714fb04eb2bf07695fd7ec053a2f32c62bc","change_number":356138,"patch_set_number":22,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2cca75d99ed8855a49a58e6343de564b931db2f5":{"kind":"REWORK","_number":5,"created":"2016-09-01 15:12:33.000000000","uploader":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"ref":"refs/changes/79/319379/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/5"}}},"commit":{"parents":[{"commit":"184a703e0671ebb18121d439b8b87dd5c0e696d8","subject":"Pull from cell0 and build_requests for instance list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/184a703e0671ebb18121d439b8b87dd5c0e696d8"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-09-01 15:12:06.000000000","tz":-240},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cca75d99ed8855a49a58e6343de564b931db2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2cca75d99ed8855a49a58e6343de564b931db2f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"184a703e0671ebb18121d439b8b87dd5c0e696d8","is_merged_in_target_branch":false,"change_id":"I625c7714fb04eb2bf07695fd7ec053a2f32c62bc","change_number":356138,"patch_set_number":22,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b965e0e8bc0059a7eba0fa7aac96b99e8d6dffa9":{"kind":"REWORK","_number":6,"created":"2016-09-06 17:58:53.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/6"}}},"commit":{"parents":[{"commit":"985c7ca4dc15176dc9cccf0ebcabaa18ea98ca2a","subject":"Merge \"Fix corrupt \"host_aggregates_map\" in host_manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/985c7ca4dc15176dc9cccf0ebcabaa18ea98ca2a"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-06 17:58:22.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b965e0e8bc0059a7eba0fa7aac96b99e8d6dffa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b965e0e8bc0059a7eba0fa7aac96b99e8d6dffa9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"985c7ca4dc15176dc9cccf0ebcabaa18ea98ca2a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b72b9f01db0015a516aecf4dde11df75ebfb3dff":{"kind":"REWORK","_number":7,"created":"2016-09-07 12:17:31.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/7"}}},"commit":{"parents":[{"commit":"5bdf06ea96b168b50b392f4edb8caa6ab6090242","subject":"Merge \"Make api-ref bug link point to nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bdf06ea96b168b50b392f4edb8caa6ab6090242"}]}],"author":{"name":"Andrew Laski","email":"andrew@lascii.com","date":"2016-03-30 19:00:25.000000000","tz":-240},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-07 12:15:49.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b72b9f01db0015a516aecf4dde11df75ebfb3dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b72b9f01db0015a516aecf4dde11df75ebfb3dff"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bdf06ea96b168b50b392f4edb8caa6ab6090242","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"154c1dd22408f03bc8c32d23fc14f462ec7493bf":{"kind":"REWORK","_number":8,"created":"2016-09-08 17:49:54.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/8"}}},"commit":{"parents":[{"commit":"91b0e00bd6d897d9ea3adacce6f15b1d5ff4f05b","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91b0e00bd6d897d9ea3adacce6f15b1d5ff4f05b"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-08 15:22:55.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-08 15:24:10.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/154c1dd22408f03bc8c32d23fc14f462ec7493bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/154c1dd22408f03bc8c32d23fc14f462ec7493bf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"91b0e00bd6d897d9ea3adacce6f15b1d5ff4f05b","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d898bc60c97a927550c313d9038c1bcd0fde0768":{"kind":"REWORK","_number":9,"created":"2016-09-09 14:54:26.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/9"}}},"commit":{"parents":[{"commit":"e3292d1ee354012b19d5c641bb7f402101c49751","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3292d1ee354012b19d5c641bb7f402101c49751"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-08 15:22:55.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-09 14:51:30.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d898bc60c97a927550c313d9038c1bcd0fde0768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d898bc60c97a927550c313d9038c1bcd0fde0768"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e3292d1ee354012b19d5c641bb7f402101c49751","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"815b3752b392a2286b5939e651203fa13b91dd37":{"kind":"REWORK","_number":10,"created":"2016-09-12 09:59:31.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/10"}}},"commit":{"parents":[{"commit":"6e5090ffd440369252cb44e9fd3128a0f586dc58","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e5090ffd440369252cb44e9fd3128a0f586dc58"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-08 15:22:55.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-12 09:45:29.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/815b3752b392a2286b5939e651203fa13b91dd37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/815b3752b392a2286b5939e651203fa13b91dd37"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6e5090ffd440369252cb44e9fd3128a0f586dc58","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d2d6137d77b54a7155cbcf9fbeb1dcbcfff64816":{"kind":"REWORK","_number":11,"created":"2016-09-15 16:31:33.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/11"}}},"commit":{"parents":[{"commit":"6e5090ffd440369252cb44e9fd3128a0f586dc58","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e5090ffd440369252cb44e9fd3128a0f586dc58"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-08 15:22:55.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-15 14:36:30.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2d6137d77b54a7155cbcf9fbeb1dcbcfff64816"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2d6137d77b54a7155cbcf9fbeb1dcbcfff64816"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6e5090ffd440369252cb44e9fd3128a0f586dc58","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"59eb03d68025f9a99684d52724db7e98feb2c67f":{"kind":"REWORK","_number":12,"created":"2016-09-19 17:33:03.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/12"}}},"commit":{"parents":[{"commit":"67613c089638107935335fb7e200de80ce2012da","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67613c089638107935335fb7e200de80ce2012da"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 17:29:27.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59eb03d68025f9a99684d52724db7e98feb2c67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59eb03d68025f9a99684d52724db7e98feb2c67f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"67613c089638107935335fb7e200de80ce2012da","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"48f966cc6661ef0a6fe0ff8d4b1a64382e462d90":{"kind":"REWORK","_number":13,"created":"2016-09-21 16:03:45.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/13"}}},"commit":{"parents":[{"commit":"a7ae1c8c3959dbce9a2f9b71fdec6b0e2508e86e","subject":"WIP Return build_requests instead of instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7ae1c8c3959dbce9a2f9b71fdec6b0e2508e86e"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-21 15:43:38.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48f966cc6661ef0a6fe0ff8d4b1a64382e462d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48f966cc6661ef0a6fe0ff8d4b1a64382e462d90"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a7ae1c8c3959dbce9a2f9b71fdec6b0e2508e86e","is_merged_in_target_branch":false,"change_id":"I60abcd4f27dc877c4e420071be77c9fdb697ad99","change_number":374267,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d8a5017ffbba820a07324844dfd7468e7aa1dba8":{"kind":"REWORK","_number":14,"created":"2016-09-22 14:32:09.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/14"}}},"commit":{"parents":[{"commit":"39e399ebe07846362f70d78b307c312d8db87101","subject":"Return build_requests instead of instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39e399ebe07846362f70d78b307c312d8db87101"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-22 14:31:32.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8a5017ffbba820a07324844dfd7468e7aa1dba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8a5017ffbba820a07324844dfd7468e7aa1dba8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"39e399ebe07846362f70d78b307c312d8db87101","is_merged_in_target_branch":false,"change_id":"I60abcd4f27dc877c4e420071be77c9fdb697ad99","change_number":374267,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e1d0612dbdf23e805c06e137e9913dabba62cb94":{"kind":"REWORK","_number":15,"created":"2016-09-26 08:58:32.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/15"}}},"commit":{"parents":[{"commit":"39f955b122455be71c560cfae60c68ea00c645f2","subject":"Return build_requests instead of instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39f955b122455be71c560cfae60c68ea00c645f2"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-26 08:18:16.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1d0612dbdf23e805c06e137e9913dabba62cb94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1d0612dbdf23e805c06e137e9913dabba62cb94"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"39f955b122455be71c560cfae60c68ea00c645f2","is_merged_in_target_branch":false,"change_id":"I60abcd4f27dc877c4e420071be77c9fdb697ad99","change_number":374267,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8caecd1c4e545199fe143cc241d464ca51d25eb3":{"kind":"REWORK","_number":16,"created":"2016-09-26 16:14:51.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/16"}}},"commit":{"parents":[{"commit":"39f955b122455be71c560cfae60c68ea00c645f2","subject":"Return build_requests instead of instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39f955b122455be71c560cfae60c68ea00c645f2"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-26 16:13:14.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8caecd1c4e545199fe143cc241d464ca51d25eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8caecd1c4e545199fe143cc241d464ca51d25eb3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"39f955b122455be71c560cfae60c68ea00c645f2","is_merged_in_target_branch":false,"change_id":"I60abcd4f27dc877c4e420071be77c9fdb697ad99","change_number":374267,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4bc5a7421fd0e8b43dcc9005e6e803a041a3f2d1":{"kind":"REWORK","_number":17,"created":"2016-09-28 10:57:11.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/17"}}},"commit":{"parents":[{"commit":"0b175381fa00ee6c537761c3242f809c0aef53fd","subject":"Return build_requests instead of instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b175381fa00ee6c537761c3242f809c0aef53fd"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-19 10:19:58.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-28 10:55:29.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bc5a7421fd0e8b43dcc9005e6e803a041a3f2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bc5a7421fd0e8b43dcc9005e6e803a041a3f2d1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0b175381fa00ee6c537761c3242f809c0aef53fd","is_merged_in_target_branch":false,"change_id":"I60abcd4f27dc877c4e420071be77c9fdb697ad99","change_number":374267,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3fc966974d8823914c99975b33a03d7321773ae7":{"kind":"REWORK","_number":18,"created":"2016-09-29 09:40:33.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/18"}}},"commit":{"parents":[{"commit":"6d3d2122f3d60b9ffda861b4366729d6a8c44b6e","subject":"Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d3d2122f3d60b9ffda861b4366729d6a8c44b6e"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 09:39:45.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fc966974d8823914c99975b33a03d7321773ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fc966974d8823914c99975b33a03d7321773ae7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6d3d2122f3d60b9ffda861b4366729d6a8c44b6e","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"c04da39ba963f6a66f8209f5cd23a3a0de5def0b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2016-09-29 09:49:28.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/19"}}},"commit":{"parents":[{"commit":"54b34c313b2655e0a60bcc79d92e6682da6d6ac7","subject":"Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/54b34c313b2655e0a60bcc79d92e6682da6d6ac7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 09:48:33.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c04da39ba963f6a66f8209f5cd23a3a0de5def0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c04da39ba963f6a66f8209f5cd23a3a0de5def0b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"54b34c313b2655e0a60bcc79d92e6682da6d6ac7","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"a9221871c3742df06c380d1392fd361c81ccdb50":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2016-10-04 14:14:45.000000000","uploader":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"ref":"refs/changes/79/319379/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/20"}}},"commit":{"parents":[{"commit":"6e4bd2800333d82cec147bf243faf2d3e2115848","subject":"WIP Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e4bd2800333d82cec147bf243faf2d3e2115848"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-10-04 14:14:08.000000000","tz":180},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9221871c3742df06c380d1392fd361c81ccdb50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9221871c3742df06c380d1392fd361c81ccdb50"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6e4bd2800333d82cec147bf243faf2d3e2115848","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":3,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"5f5bd35393d7250878e77af3398468c29ade45d4":{"kind":"REWORK","_number":21,"created":"2016-11-02 14:28:50.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/21"}}},"commit":{"parents":[{"commit":"624852d1468b54a44e548f5c689e4243dfac4c77","subject":"WIP Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/624852d1468b54a44e548f5c689e4243dfac4c77"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-02 14:33:52.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5bd35393d7250878e77af3398468c29ade45d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5bd35393d7250878e77af3398468c29ade45d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"624852d1468b54a44e548f5c689e4243dfac4c77","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":4,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"f70cb2afdab9f3395c346c976cc05ea89534d44d":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2016-11-02 18:23:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/22"}}},"commit":{"parents":[{"commit":"7cbcaf8a9f3170835d25807e0dced44271dc24f9","subject":"WIP Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7cbcaf8a9f3170835d25807e0dced44271dc24f9"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-02 18:29:12.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f70cb2afdab9f3395c346c976cc05ea89534d44d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f70cb2afdab9f3395c346c976cc05ea89534d44d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7cbcaf8a9f3170835d25807e0dced44271dc24f9","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":5,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"ebf1b23356e55b0a3118ca267e97c9686652d9e2":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2016-11-02 20:03:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/23"}}},"commit":{"parents":[{"commit":"1740f427e0af62c5db5e53983f6dce4846a3cb78","subject":"WIP Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1740f427e0af62c5db5e53983f6dce4846a3cb78"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-02 20:09:36.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ebf1b23356e55b0a3118ca267e97c9686652d9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ebf1b23356e55b0a3118ca267e97c9686652d9e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1740f427e0af62c5db5e53983f6dce4846a3cb78","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":6,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"abd79b3ce23b7587bc4e357d317c1967ac72ebb1":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2016-11-02 20:36:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/24"}}},"commit":{"parents":[{"commit":"0f95628f070dd30553402ee8c3978b54b29e67cc","subject":"WIP Add schedule_and_build_instances RPC method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f95628f070dd30553402ee8c3978b54b29e67cc"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-02 20:42:27.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abd79b3ce23b7587bc4e357d317c1967ac72ebb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abd79b3ce23b7587bc4e357d317c1967ac72ebb1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0f95628f070dd30553402ee8c3978b54b29e67cc","is_merged_in_target_branch":false,"change_id":"I04db4ec98a862eeb24d1ed9f54774c39f4617632","change_number":379320,"patch_set_number":7,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"c0264170575080f7278930333cb3a62da0e8154e":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2016-11-03 14:01:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/25"}}},"commit":{"parents":[{"commit":"5e24ff21e5e4d4926194e03ed326c7c3cb00f77d","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e24ff21e5e4d4926194e03ed326c7c3cb00f77d"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-03 14:07:25.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0264170575080f7278930333cb3a62da0e8154e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0264170575080f7278930333cb3a62da0e8154e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e24ff21e5e4d4926194e03ed326c7c3cb00f77d","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2f8486e67f3640f493293cba93a54b0d954cf40f":{"kind":"REWORK","_number":26,"created":"2016-11-03 18:18:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/26"}}},"commit":{"parents":[{"commit":"e49fb18682d5bada38774974db04ea3d9c9b9e5b","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e49fb18682d5bada38774974db04ea3d9c9b9e5b"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-03 18:23:27.000000000","tz":-420},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f8486e67f3640f493293cba93a54b0d954cf40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f8486e67f3640f493293cba93a54b0d954cf40f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e49fb18682d5bada38774974db04ea3d9c9b9e5b","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bcc4361458e17567de738becd1a9952ebd1cd9fb":{"kind":"REWORK","_number":27,"created":"2016-11-07 22:13:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/27"}}},"commit":{"parents":[{"commit":"b4f9062e66096f9fc1f222313c04ae3b8f3bf47c","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4f9062e66096f9fc1f222313c04ae3b8f3bf47c"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-07 22:18:49.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bcc4361458e17567de738becd1a9952ebd1cd9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bcc4361458e17567de738becd1a9952ebd1cd9fb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b4f9062e66096f9fc1f222313c04ae3b8f3bf47c","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"280b0ac3362e294df9d663e12e74dc4ea1fc9f58":{"kind":"REWORK","_number":28,"created":"2016-11-08 15:57:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/28"}}},"commit":{"parents":[{"commit":"b4f9062e66096f9fc1f222313c04ae3b8f3bf47c","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4f9062e66096f9fc1f222313c04ae3b8f3bf47c"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-08 16:03:00.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/280b0ac3362e294df9d663e12e74dc4ea1fc9f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/280b0ac3362e294df9d663e12e74dc4ea1fc9f58"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b4f9062e66096f9fc1f222313c04ae3b8f3bf47c","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a03a608f3cb31bdb108789bee710d920bd6684cd":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2016-11-08 17:30:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/29"}}},"commit":{"parents":[{"commit":"b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-08 17:36:35.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a03a608f3cb31bdb108789bee710d920bd6684cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a03a608f3cb31bdb108789bee710d920bd6684cd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0dc3051053044feb97ed29ae429b163bd22278bb":{"kind":"REWORK","_number":30,"created":"2016-11-08 21:05:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/30"}}},"commit":{"parents":[{"commit":"b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-08 21:11:23.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dc3051053044feb97ed29ae429b163bd22278bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dc3051053044feb97ed29ae429b163bd22278bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b2c111c8d47ed3c7ece0a6ef8e582ac5ba38e18d","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ad3d15ae76410f58bad589756e3e5107b58c411b":{"kind":"REWORK","_number":31,"created":"2016-11-10 16:38:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/31"}}},"commit":{"parents":[{"commit":"acb1051824af14493b8e6191e8c576e51ae72d47","subject":"Refactor two nearly useless secgroup tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/acb1051824af14493b8e6191e8c576e51ae72d47"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-10 16:44:08.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad3d15ae76410f58bad589756e3e5107b58c411b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad3d15ae76410f58bad589756e3e5107b58c411b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"acb1051824af14493b8e6191e8c576e51ae72d47","is_merged_in_target_branch":false,"change_id":"I58f2d9d166446cd762a2fd4ad9d6b4bf5fff0c1d","change_number":393438,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a9b2f3b873c08b3842df0d9ce530abd4bf57f45b":{"kind":"REWORK","_number":32,"created":"2016-11-10 21:57:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/32"}}},"commit":{"parents":[{"commit":"7d458ba30141c1a2d4179558bf123c6ce07ffb93","subject":"Make RPCFixture support multiple connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d458ba30141c1a2d4179558bf123c6ce07ffb93"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-10 22:03:46.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9b2f3b873c08b3842df0d9ce530abd4bf57f45b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9b2f3b873c08b3842df0d9ce530abd4bf57f45b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7d458ba30141c1a2d4179558bf123c6ce07ffb93","is_merged_in_target_branch":false,"change_id":"Icb63d7dabd17f3c5633387793f68a8ba20863a7e","change_number":396417,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"35012c6b83f8e7893f0b16a3c314662349261692":{"kind":"REWORK","_number":33,"created":"2016-11-11 21:01:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/33"}}},"commit":{"parents":[{"commit":"e9854a263aac67a603a948ebfdeb1f9c7c6d09d4","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e9854a263aac67a603a948ebfdeb1f9c7c6d09d4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-11 21:07:05.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35012c6b83f8e7893f0b16a3c314662349261692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35012c6b83f8e7893f0b16a3c314662349261692"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e9854a263aac67a603a948ebfdeb1f9c7c6d09d4","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"819c6f09291b13fdf18c5f733594bea47b827230":{"kind":"REWORK","_number":34,"created":"2016-11-30 20:17:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/34"}}},"commit":{"parents":[{"commit":"d0e9c95bf9283ac7e8c4c92b050daa6c3a6fc8e6","subject":"Add schedule_and_build_instances conductor method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0e9c95bf9283ac7e8c4c92b050daa6c3a6fc8e6"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-30 20:23:32.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/819c6f09291b13fdf18c5f733594bea47b827230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/819c6f09291b13fdf18c5f733594bea47b827230"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d0e9c95bf9283ac7e8c4c92b050daa6c3a6fc8e6","is_merged_in_target_branch":false,"change_id":"I19d6d765894785b214458b32fe612b4ef78a270d","change_number":367557,"patch_set_number":21,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c231a413b0bbccf22a6420c3f1623549c0099be4":{"kind":"REWORK","_number":35,"created":"2016-11-30 23:41:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/35"}}},"commit":{"parents":[{"commit":"5d67132afa7c141dde1dc9bf5c1e7530fb4e3406","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d67132afa7c141dde1dc9bf5c1e7530fb4e3406"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-11-30 23:46:21.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c231a413b0bbccf22a6420c3f1623549c0099be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c231a413b0bbccf22a6420c3f1623549c0099be4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5d67132afa7c141dde1dc9bf5c1e7530fb4e3406","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ea2d8572e1dcb9f64d9e43eb8792c2e0f24e4c10":{"kind":"REWORK","_number":36,"created":"2016-12-02 00:25:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/36"}}},"commit":{"parents":[{"commit":"aace934f8e21183491a8db8ac27ef98717ae4839","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aace934f8e21183491a8db8ac27ef98717ae4839"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-01 17:46:24.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea2d8572e1dcb9f64d9e43eb8792c2e0f24e4c10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea2d8572e1dcb9f64d9e43eb8792c2e0f24e4c10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aace934f8e21183491a8db8ac27ef98717ae4839","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7ee964ba9b118be3df343f466f9385b3b99d5fbf":{"kind":"REWORK","_number":37,"created":"2016-12-02 17:27:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/37"}}},"commit":{"parents":[{"commit":"1cc4f3223b18dadad3374df4415ca25a3ad83bb8","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cc4f3223b18dadad3374df4415ca25a3ad83bb8"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-02 17:34:03.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ee964ba9b118be3df343f466f9385b3b99d5fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ee964ba9b118be3df343f466f9385b3b99d5fbf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1cc4f3223b18dadad3374df4415ca25a3ad83bb8","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9b57a593e45cc2ce6bf935ee7bc33b8118889eb8":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2016-12-02 22:26:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/38"}}},"commit":{"parents":[{"commit":"eeb0eef021af23df95345da224780b88afcd97fc","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eeb0eef021af23df95345da224780b88afcd97fc"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-02 22:32:01.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b57a593e45cc2ce6bf935ee7bc33b8118889eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b57a593e45cc2ce6bf935ee7bc33b8118889eb8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eeb0eef021af23df95345da224780b88afcd97fc","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"726547dfcab4d2e25a6df82ddcef56ec2e97b876":{"kind":"REWORK","_number":39,"created":"2016-12-03 01:20:54.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/39"}}},"commit":{"parents":[{"commit":"47053e12bd201137bf4cc9aeb4d8a619e1ce30cf","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/47053e12bd201137bf4cc9aeb4d8a619e1ce30cf"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-03 01:26:51.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/726547dfcab4d2e25a6df82ddcef56ec2e97b876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/726547dfcab4d2e25a6df82ddcef56ec2e97b876"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"47053e12bd201137bf4cc9aeb4d8a619e1ce30cf","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"430bb0d65d2acfd11e01e86f25d77f2420734234":{"kind":"REWORK","_number":40,"created":"2016-12-05 16:29:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/40"}}},"commit":{"parents":[{"commit":"5e183d687d63c5e3091e29caf6efa57bed3c5474","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e183d687d63c5e3091e29caf6efa57bed3c5474"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-05 16:35:41.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/430bb0d65d2acfd11e01e86f25d77f2420734234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/430bb0d65d2acfd11e01e86f25d77f2420734234"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e183d687d63c5e3091e29caf6efa57bed3c5474","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3217d267900fcada8c79d23d7ddf08d0f12435a4":{"kind":"REWORK","_number":41,"created":"2016-12-05 23:42:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/41"}}},"commit":{"parents":[{"commit":"5dec6c578c0f49784a69e25a486ac38cb5df6f91","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5dec6c578c0f49784a69e25a486ac38cb5df6f91"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-05 23:48:29.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3217d267900fcada8c79d23d7ddf08d0f12435a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3217d267900fcada8c79d23d7ddf08d0f12435a4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5dec6c578c0f49784a69e25a486ac38cb5df6f91","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"503f6e2f23a0698c7471c155cdda184387fc6143":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2016-12-06 18:17:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/42"}}},"commit":{"parents":[{"commit":"12254dabd5e8f3d6315362b1ad1eb7ed7756df72","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12254dabd5e8f3d6315362b1ad1eb7ed7756df72"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-06 18:20:50.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/503f6e2f23a0698c7471c155cdda184387fc6143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/503f6e2f23a0698c7471c155cdda184387fc6143"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12254dabd5e8f3d6315362b1ad1eb7ed7756df72","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e7151de7876553406cf4ccf1a98c505d13adac69":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2016-12-07 00:24:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/43"}}},"commit":{"parents":[{"commit":"8aad9c3574ab98e239ab2a9686b9899e72c9a840","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8aad9c3574ab98e239ab2a9686b9899e72c9a840"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-07 00:30:55.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7151de7876553406cf4ccf1a98c505d13adac69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7151de7876553406cf4ccf1a98c505d13adac69"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8aad9c3574ab98e239ab2a9686b9899e72c9a840","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7daf897dd827fd007f7721dbd9f8747d5b2c5023":{"kind":"REWORK","_number":44,"created":"2016-12-07 14:45:37.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/44"}}},"commit":{"parents":[{"commit":"89ea03f0031c4360149862d64980d2d6757bb738","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89ea03f0031c4360149862d64980d2d6757bb738"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-07 14:51:55.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7daf897dd827fd007f7721dbd9f8747d5b2c5023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7daf897dd827fd007f7721dbd9f8747d5b2c5023"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"89ea03f0031c4360149862d64980d2d6757bb738","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b5a42e038c1ae8887fb7cae99c41eeb403e0ed4a":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2016-12-07 14:47:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/45"}}},"commit":{"parents":[{"commit":"d102ae166e5f1217c49938e5c2a24146c6d10941","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d102ae166e5f1217c49938e5c2a24146c6d10941"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-07 14:54:15.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5a42e038c1ae8887fb7cae99c41eeb403e0ed4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5a42e038c1ae8887fb7cae99c41eeb403e0ed4a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d102ae166e5f1217c49938e5c2a24146c6d10941","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b2be90ad18c50178c8ece8aa15e99c0e66fbfc27":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2016-12-07 18:17:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/46"}}},"commit":{"parents":[{"commit":"7393d52a30513502249db753cce0aa69d952c0ae","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7393d52a30513502249db753cce0aa69d952c0ae"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-07 18:24:07.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2be90ad18c50178c8ece8aa15e99c0e66fbfc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2be90ad18c50178c8ece8aa15e99c0e66fbfc27"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7393d52a30513502249db753cce0aa69d952c0ae","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e8ce56198471b72f140145d67d096686b8778998":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2016-12-09 18:25:06.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/47"}}},"commit":{"parents":[{"commit":"5efeb664e3ae3d115e89731b17f6e7e96da0d6fd","subject":"WIP Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5efeb664e3ae3d115e89731b17f6e7e96da0d6fd"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-09 18:30:48.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e8ce56198471b72f140145d67d096686b8778998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e8ce56198471b72f140145d67d096686b8778998"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5efeb664e3ae3d115e89731b17f6e7e96da0d6fd","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"96398dd84feae5816c9220a6dc9a11cfa039cfb6":{"kind":"REWORK","_number":48,"created":"2016-12-14 00:18:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/48"}}},"commit":{"parents":[{"commit":"4b9c524a5f21268d572e0494cb719df0c723465d","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b9c524a5f21268d572e0494cb719df0c723465d"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-13 23:29:24.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96398dd84feae5816c9220a6dc9a11cfa039cfb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96398dd84feae5816c9220a6dc9a11cfa039cfb6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4b9c524a5f21268d572e0494cb719df0c723465d","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bed911a303349d68509cc35f1ecefcc63d5a9ea3":{"kind":"REWORK","_number":49,"created":"2016-12-14 19:50:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/49"}}},"commit":{"parents":[{"commit":"1a4d0b3789352e4ce40f303737faa7b83797081f","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a4d0b3789352e4ce40f303737faa7b83797081f"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-12-14 19:57:00.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bed911a303349d68509cc35f1ecefcc63d5a9ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bed911a303349d68509cc35f1ecefcc63d5a9ea3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1a4d0b3789352e4ce40f303737faa7b83797081f","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f45bf1f893fa9dd89790ba94d4e76e9abf5a8336":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2017-01-03 19:41:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/50"}}},"commit":{"parents":[{"commit":"8425db07b86b05d3501536389caad7a109ab8826","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8425db07b86b05d3501536389caad7a109ab8826"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-03 19:48:11.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f45bf1f893fa9dd89790ba94d4e76e9abf5a8336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f45bf1f893fa9dd89790ba94d4e76e9abf5a8336"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8425db07b86b05d3501536389caad7a109ab8826","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"925cbc9e46a2bc3e0356954bfadde40cead7ce25":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2017-01-03 21:35:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/51"}}},"commit":{"parents":[{"commit":"bc064522970bf5a252133498f9f8e44813199ca3","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc064522970bf5a252133498f9f8e44813199ca3"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-03 21:42:32.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/925cbc9e46a2bc3e0356954bfadde40cead7ce25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/925cbc9e46a2bc3e0356954bfadde40cead7ce25"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bc064522970bf5a252133498f9f8e44813199ca3","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"59f5230d3da1729a39c36db335342d19f9aec675":{"kind":"REWORK","_number":52,"created":"2017-01-04 18:10:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/52"}}},"commit":{"parents":[{"commit":"d2d66c400d94b254bb882c92acbf1df43cb98af7","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2d66c400d94b254bb882c92acbf1df43cb98af7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-04 18:17:54.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59f5230d3da1729a39c36db335342d19f9aec675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59f5230d3da1729a39c36db335342d19f9aec675"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d2d66c400d94b254bb882c92acbf1df43cb98af7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"55e9f6c1e0e9138a204b855bf452b86937bfc9dc":{"kind":"NO_CHANGE","_number":53,"created":"2017-01-04 23:30:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/53"}}},"commit":{"parents":[{"commit":"589b7f721ac489f96d60ef48ce36ee56797e40dc","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/589b7f721ac489f96d60ef48ce36ee56797e40dc"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-04 23:03:39.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55e9f6c1e0e9138a204b855bf452b86937bfc9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55e9f6c1e0e9138a204b855bf452b86937bfc9dc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"589b7f721ac489f96d60ef48ce36ee56797e40dc","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c6e14849cd9ef4682800fe7cfb7a09e174f9ecdd":{"kind":"REWORK","_number":54,"created":"2017-01-12 15:37:48.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/54"}}},"commit":{"parents":[{"commit":"e74e3ce038a4a5c7788fbf169b6c1c0356374c78","subject":"Multicell support for instance listing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e74e3ce038a4a5c7788fbf169b6c1c0356374c78"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-12 15:44:58.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6e14849cd9ef4682800fe7cfb7a09e174f9ecdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6e14849cd9ef4682800fe7cfb7a09e174f9ecdd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e74e3ce038a4a5c7788fbf169b6c1c0356374c78","is_merged_in_target_branch":false,"change_id":"I37b5af1e02e2288340378944e77c110268a59a8d","change_number":396775,"patch_set_number":37,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0096e331ae769f6b22a5979a0e9f4b6bec4151b6":{"kind":"REWORK","_number":55,"created":"2017-01-12 18:30:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/55"}}},"commit":{"parents":[{"commit":"a543c06cc3793cb57bb9909c632cf5a69e4055ee","subject":"WIP: Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a543c06cc3793cb57bb9909c632cf5a69e4055ee"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-12 18:19:42.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0096e331ae769f6b22a5979a0e9f4b6bec4151b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0096e331ae769f6b22a5979a0e9f4b6bec4151b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a543c06cc3793cb57bb9909c632cf5a69e4055ee","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ce1ab5b6d352127e1c17aa56f20205bac5906e7b":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2017-01-12 22:21:16.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/56"}}},"commit":{"parents":[{"commit":"6ac872bbf8eed09e342a20b1626084bbf6ce2f08","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ac872bbf8eed09e342a20b1626084bbf6ce2f08"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-12 22:25:36.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce1ab5b6d352127e1c17aa56f20205bac5906e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce1ab5b6d352127e1c17aa56f20205bac5906e7b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6ac872bbf8eed09e342a20b1626084bbf6ce2f08","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3ff7818403b002e4e23c90c48db6af32544142e1":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2017-01-13 20:48:18.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/79/319379/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/57"}}},"commit":{"parents":[{"commit":"df5df1d9280af566b9df99fb17bdf583db6a11f3","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df5df1d9280af566b9df99fb17bdf583db6a11f3"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2017-01-13 20:48:05.000000000","tz":-300},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ff7818403b002e4e23c90c48db6af32544142e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ff7818403b002e4e23c90c48db6af32544142e1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"df5df1d9280af566b9df99fb17bdf583db6a11f3","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"eeb8ee38619a7d0bd6bfda035a17d8a07576e547":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2017-01-14 19:58:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/58"}}},"commit":{"parents":[{"commit":"6316e9182c738f00ee19ec562bc44835529f6713","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6316e9182c738f00ee19ec562bc44835529f6713"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-14 20:05:10.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eeb8ee38619a7d0bd6bfda035a17d8a07576e547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eeb8ee38619a7d0bd6bfda035a17d8a07576e547"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6316e9182c738f00ee19ec562bc44835529f6713","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4b62f46ba4d9ad693575b1637aacb8330bc577d2":{"kind":"REWORK","_number":59,"created":"2017-01-16 21:24:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/59"}}},"commit":{"parents":[{"commit":"6316e9182c738f00ee19ec562bc44835529f6713","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6316e9182c738f00ee19ec562bc44835529f6713"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-16 21:31:41.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b62f46ba4d9ad693575b1637aacb8330bc577d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b62f46ba4d9ad693575b1637aacb8330bc577d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6316e9182c738f00ee19ec562bc44835529f6713","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f9eed05271e5a4ceb92e42f777a2c59919213702":{"kind":"REWORK","_number":60,"created":"2017-01-16 22:21:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/60"}}},"commit":{"parents":[{"commit":"6316e9182c738f00ee19ec562bc44835529f6713","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6316e9182c738f00ee19ec562bc44835529f6713"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-16 22:28:31.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: If1af9c478e8ea2420f2523a9bb8b70fafddc86b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9eed05271e5a4ceb92e42f777a2c59919213702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9eed05271e5a4ceb92e42f777a2c59919213702"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6316e9182c738f00ee19ec562bc44835529f6713","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dd3a50733cdb3b64f1677107b8f58b2312bd4c4b":{"kind":"REWORK","_number":61,"created":"2017-01-16 23:07:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/61"}}},"commit":{"parents":[{"commit":"6316e9182c738f00ee19ec562bc44835529f6713","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6316e9182c738f00ee19ec562bc44835529f6713"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-16 23:15:10.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd3a50733cdb3b64f1677107b8f58b2312bd4c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd3a50733cdb3b64f1677107b8f58b2312bd4c4b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6316e9182c738f00ee19ec562bc44835529f6713","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ee4c21e6b01fc2aca4f131f2c204459c671cb023":{"kind":"REWORK","_number":62,"created":"2017-01-16 23:44:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/62"}}},"commit":{"parents":[{"commit":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-16 23:52:14.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee4c21e6b01fc2aca4f131f2c204459c671cb023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee4c21e6b01fc2aca4f131f2c204459c671cb023"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9830d460e0e46a9c59a046b1d5a64b02c8878219":{"kind":"REWORK","_number":63,"created":"2017-01-17 02:20:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/63"}}},"commit":{"parents":[{"commit":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-17 02:27:51.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9830d460e0e46a9c59a046b1d5a64b02c8878219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9830d460e0e46a9c59a046b1d5a64b02c8878219"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"abd37363f5800f5ad84a1d14526da18e3d665ebb":{"kind":"REWORK","_number":64,"created":"2017-01-20 02:26:47.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/79/319379/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/64"}}},"commit":{"parents":[{"commit":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-01-20 01:42:05.000000000","tz":0},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abd37363f5800f5ad84a1d14526da18e3d665ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abd37363f5800f5ad84a1d14526da18e3d665ebb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"061fe281f98a767049ffe3753801f52759f31229":{"kind":"REWORK","_number":66,"created":"2017-01-22 00:24:58.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/79/319379/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/66"}}},"commit":{"parents":[{"commit":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-01-22 00:24:06.000000000","tz":0},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/061fe281f98a767049ffe3753801f52759f31229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/061fe281f98a767049ffe3753801f52759f31229"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d611e6790ea5a25cf0680d744e19912c1d4a9501":{"kind":"REWORK","_number":67,"created":"2017-01-22 01:46:00.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/79/319379/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/67"}}},"commit":{"parents":[{"commit":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-01-22 01:43:28.000000000","tz":0},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d611e6790ea5a25cf0680d744e19912c1d4a9501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d611e6790ea5a25cf0680d744e19912c1d4a9501"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51504ce4aaf0e7ddcd4ad66dbb170e25440e2ef7","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"86447a2aa4d3197aee5cf97fd233aa531e29b8e3":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2017-01-23 16:35:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/68"}}},"commit":{"parents":[{"commit":"9735fe5a34452ccb8012d806934e94eea4aaeb0f","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9735fe5a34452ccb8012d806934e94eea4aaeb0f"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-23 16:41:27.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: Ic6e11dadec4827ade89b1c0019f8d1493a9fe96e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/86447a2aa4d3197aee5cf97fd233aa531e29b8e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/86447a2aa4d3197aee5cf97fd233aa531e29b8e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9735fe5a34452ccb8012d806934e94eea4aaeb0f","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fa28439f679979c0469f7ef9b8dff9b15b4843f9":{"kind":"NO_CODE_CHANGE","_number":69,"created":"2017-01-23 17:26:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/69"}}},"commit":{"parents":[{"commit":"9735fe5a34452ccb8012d806934e94eea4aaeb0f","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9735fe5a34452ccb8012d806934e94eea4aaeb0f"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-23 17:33:47.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa28439f679979c0469f7ef9b8dff9b15b4843f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa28439f679979c0469f7ef9b8dff9b15b4843f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9735fe5a34452ccb8012d806934e94eea4aaeb0f","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f66fcc53820d0515b95bd333f522c9e5dafb442d":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2017-01-23 18:17:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/70"}}},"commit":{"parents":[{"commit":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea65c96abaf350f97a5fe4967906d6b52ef13d4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-23 18:24:39.000000000","tz":-480},"subject":"WIP Move instance creation to conductor","message":"WIP Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f66fcc53820d0515b95bd333f522c9e5dafb442d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f66fcc53820d0515b95bd333f522c9e5dafb442d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1ffb554a5458abaeec8ae1e88546e0b7040bc4e9":{"kind":"NO_CODE_CHANGE","_number":71,"created":"2017-01-24 00:56:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/71"}}},"commit":{"parents":[{"commit":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea65c96abaf350f97a5fe4967906d6b52ef13d4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 00:52:56.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ffb554a5458abaeec8ae1e88546e0b7040bc4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ffb554a5458abaeec8ae1e88546e0b7040bc4e9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2779dbf323d0db33e60be431aa7ec7f250d48990":{"kind":"REWORK","_number":72,"created":"2017-01-24 16:07:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/72"}}},"commit":{"parents":[{"commit":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea65c96abaf350f97a5fe4967906d6b52ef13d4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 16:10:27.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2779dbf323d0db33e60be431aa7ec7f250d48990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2779dbf323d0db33e60be431aa7ec7f250d48990"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d4ae4419adbd702191f30860c7e2c9df89ed0c0f":{"kind":"REWORK","_number":73,"created":"2017-01-24 16:36:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/73"}}},"commit":{"parents":[{"commit":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea65c96abaf350f97a5fe4967906d6b52ef13d4"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 16:43:38.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4ae4419adbd702191f30860c7e2c9df89ed0c0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4ae4419adbd702191f30860c7e2c9df89ed0c0f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ea65c96abaf350f97a5fe4967906d6b52ef13d4","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"88014b629be9730a8e925686dccec2832b3b549b":{"kind":"REWORK","_number":74,"created":"2017-01-24 17:36:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/74"}}},"commit":{"parents":[{"commit":"ecc06e7b2ac66e95a3fcb7358ea70aec2823cabd","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ecc06e7b2ac66e95a3fcb7358ea70aec2823cabd"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 17:43:37.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88014b629be9730a8e925686dccec2832b3b549b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88014b629be9730a8e925686dccec2832b3b549b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ecc06e7b2ac66e95a3fcb7358ea70aec2823cabd","is_merged_in_target_branch":false,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dffdb3ce169f352bc12516963e853e3930035396":{"kind":"TRIVIAL_REBASE","_number":75,"created":"2017-01-24 17:52:42.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/75"}}},"commit":{"parents":[{"commit":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e3b2abdc634ccd1f2e897752da11f1b70ba7aac"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 17:59:55.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dffdb3ce169f352bc12516963e853e3930035396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dffdb3ce169f352bc12516963e853e3930035396"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","is_merged_in_target_branch":true,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6143cf0c55940a1170c5bb27dceb0d49bbe7399a":{"kind":"REWORK","_number":76,"created":"2017-01-24 19:20:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/76"}}},"commit":{"parents":[{"commit":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e3b2abdc634ccd1f2e897752da11f1b70ba7aac"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 19:27:43.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6143cf0c55940a1170c5bb27dceb0d49bbe7399a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6143cf0c55940a1170c5bb27dceb0d49bbe7399a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","is_merged_in_target_branch":true,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"09e4325c00dd39c4d9dc6da94e546538a263067a":{"kind":"REWORK","_number":77,"created":"2017-01-24 21:01:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/77"}}},"commit":{"parents":[{"commit":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","subject":"Make api_samples tests use simple cell environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e3b2abdc634ccd1f2e897752da11f1b70ba7aac"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 21:03:48.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09e4325c00dd39c4d9dc6da94e546538a263067a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09e4325c00dd39c4d9dc6da94e546538a263067a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4e3b2abdc634ccd1f2e897752da11f1b70ba7aac","is_merged_in_target_branch":true,"change_id":"If1138331f3a46f5aed87e898ce19879a787d435f","change_number":410473,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2bcf7aaddc7d24381c0c22a1628ddd77385ca198":{"kind":"REWORK","_number":78,"created":"2017-01-24 23:43:05.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/78"}}},"commit":{"parents":[{"commit":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","subject":"Merge \"Make api_samples tests use simple cell environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20cf491971bbedbaa2358aa00b0f9afbe46d3ff1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-24 23:49:43.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bcf7aaddc7d24381c0c22a1628ddd77385ca198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bcf7aaddc7d24381c0c22a1628ddd77385ca198"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7223d163d5ee945ff10d5257e1602e93f173e982":{"kind":"REWORK","_number":79,"created":"2017-01-25 15:17:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/79"}}},"commit":{"parents":[{"commit":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","subject":"Merge \"Make api_samples tests use simple cell environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20cf491971bbedbaa2358aa00b0f9afbe46d3ff1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-25 15:24:45.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7223d163d5ee945ff10d5257e1602e93f173e982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7223d163d5ee945ff10d5257e1602e93f173e982"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"691459ab7500a7eed90dee485cb9393da35c848c":{"kind":"REWORK","_number":80,"created":"2017-01-25 21:39:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/80"}}},"commit":{"parents":[{"commit":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","subject":"Merge \"Make api_samples tests use simple cell environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20cf491971bbedbaa2358aa00b0f9afbe46d3ff1"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-25 21:22:47.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/691459ab7500a7eed90dee485cb9393da35c848c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/691459ab7500a7eed90dee485cb9393da35c848c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20cf491971bbedbaa2358aa00b0f9afbe46d3ff1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7ab7ae2608a37e3a75794ec86d3df176bb58568b":{"kind":"REWORK","_number":81,"created":"2017-01-25 23:36:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/81"}}},"commit":{"parents":[{"commit":"14daabd30845966f5a85bad4ef1d961f46e4098c","subject":"Merge \"libvirt: avoid generating script with empty path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14daabd30845966f5a85bad4ef1d961f46e4098c"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-25 23:43:08.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ab7ae2608a37e3a75794ec86d3df176bb58568b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ab7ae2608a37e3a75794ec86d3df176bb58568b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"14daabd30845966f5a85bad4ef1d961f46e4098c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4954d9abfa5495ed6f29b3aa3941970f98386a22":{"kind":"REWORK","_number":82,"created":"2017-01-26 03:02:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/82"}}},"commit":{"parents":[{"commit":"14daabd30845966f5a85bad4ef1d961f46e4098c","subject":"Merge \"libvirt: avoid generating script with empty path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14daabd30845966f5a85bad4ef1d961f46e4098c"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-26 03:09:21.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4954d9abfa5495ed6f29b3aa3941970f98386a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4954d9abfa5495ed6f29b3aa3941970f98386a22"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"14daabd30845966f5a85bad4ef1d961f46e4098c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a095d5d00ec29575997289d79e25f757cf4f92fd":{"kind":"REWORK","_number":83,"created":"2017-01-26 06:52:42.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/79/319379/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/83"}}},"commit":{"parents":[{"commit":"14daabd30845966f5a85bad4ef1d961f46e4098c","subject":"Merge \"libvirt: avoid generating script with empty path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14daabd30845966f5a85bad4ef1d961f46e4098c"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2017-01-26 06:52:07.000000000","tz":0},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a095d5d00ec29575997289d79e25f757cf4f92fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a095d5d00ec29575997289d79e25f757cf4f92fd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"14daabd30845966f5a85bad4ef1d961f46e4098c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3c2ce9450bea8d417bfd3fbe26ee688446c6a348":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2017-01-26 19:56:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/79/319379/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/319379/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/319379/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/319379/84"}}},"commit":{"parents":[{"commit":"f9d7b383a7cb12b6cd3e6117daf69b08620bf40f","subject":"Merge \"PCI: Check pci_requests object is empty before passing to support_requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9d7b383a7cb12b6cd3e6117daf69b08620bf40f"}]}],"author":{"name":"Andrey Volkov","email":"avolkov@mirantis.com","date":"2016-09-29 08:43:09.000000000","tz":180},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2017-01-26 20:02:54.000000000","tz":-480},"subject":"Move instance creation to conductor","message":"Move instance creation to conductor\n\nThis reworks a big chunk of the instance boot flow. A new method has\nbeen added to the conductor which will handle the initial scheduling and\ncreation of an instance in the cell db. At that point the request moves\nto the compute as normal. A reschedule will call back to the cell\nconductor as normal and can not result in a host from another cell being\nselected.\n\nCo-Authored-By: Dan Smith \u003cdansmith@redhat.com\u003e\nImplements: bp cells-scheduling-interaction\nChange-Id: I8742071b55f018f864f5a382de20075a5b444a79\nDepends-On: I8938fd37f71498e87aa348a81c357eef5c1a8093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c2ce9450bea8d417bfd3fbe26ee688446c6a348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c2ce9450bea8d417bfd3fbe26ee688446c6a348"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9d7b383a7cb12b6cd3e6117daf69b08620bf40f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
