)]}'
{"id":"openstack%2Fnova~325462","triplet_id":"openstack%2Fnova~master~I79d5d82d32c157853de112e815089e71e16b39d3","project":"openstack/nova","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I79d5d82d32c157853de112e815089e71e16b39d3","subject":"Updated from global requirements","status":"MERGED","created":"2016-06-03 18:17:19.000000000","updated":"2016-06-07 20:53:17.000000000","submitted":"2016-06-07 20:53:15.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"205ab828330d4cb7dc7383940caf54268c3e5c82","_number":325462,"virtual_id_number":325462,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-07 20:53:15.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":1,"date":"2016-06-06 04:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-06 15:49:51.000000000","_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2016-06-06 15:49:51.000000000","_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-03 19:35:31.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-03 19:40:31.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-03 19:47:33.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 04:45:09.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-06 15:49:51.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"},{"updated":"2016-06-07 20:53:15.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"57c42a8870dfac6d9d329f47a9128d4c5857f4ef","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-06-03 18:17:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dbb0774ecf8a845ff06bef6074a39a32302425b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-03 18:38:33.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/62/325462/1/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/4c918544616b447697d54449b14a643d/index.html : SUCCESS in 17m 57s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/62/325462/1/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/76b481e51bd7401ba508e0b75f33624d/index.html : SUCCESS in 15m 46s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/62/325462/1/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e592266832444192b1d1c7cd9a1a3195/index.html : SUCCESS in 19m 43s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/62/325462/1/check/real-db-upgrade_nova_percona_user_001:th-percona/d41507f8687f40c3ac6a38940c292d8c/index.html : SUCCESS in 16m 03s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":1},{"id":"9295f1095ef6915fc819961ca852739cb3e5db2d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-03 19:35:31.000000000","message":"Patch Set 1: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/62/325462/1/17527/index.html : SUCCESS in 01 h 15 m 48 s","accounts_in_message":[],"_revision_number":1},{"id":"3f8cad006597bde4c3ba48bbb00a82b8033207b7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-06-03 19:40:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/62/325462/1/check/dsvm-tempest-xen/4047ed0 : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"db652edfbca5de8c00acaf8d6ce8205bcfa86e83","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-03 19:47:33.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/62/325462/1/check/check-ibm-tempest-dsvm-full/9c95770/ : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f0958166f90e66065d1a12db5e5f1224da3d1fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2016-06-03 20:13:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f9813403380e94f43495fa52f4b381a36d398e77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-04 02:34:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/62/325462/1/check/gate-nova-docs/4a8f215//doc/build/html/ : SUCCESS in 1m 20s\n- gate-nova-pep8 http://logs.openstack.org/62/325462/1/check/gate-nova-pep8/95f546a/ : SUCCESS in 5m 03s\n- gate-nova-python27-db http://logs.openstack.org/62/325462/1/check/gate-nova-python27-db/77d6b7a/ : SUCCESS in 8m 20s\n- gate-nova-python34-db http://logs.openstack.org/62/325462/1/check/gate-nova-python34-db/6487e2c/ : SUCCESS in 10m 20s\n- gate-nova-requirements http://logs.openstack.org/62/325462/1/check/gate-nova-requirements/a682360/ : SUCCESS in 1m 59s\n- gate-tempest-dsvm-full http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-full/c271fd4/ : SUCCESS in 45m 28s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-postgres-full/27c9b09/ : SUCCESS in 52m 29s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-neutron-full/112b955/ : SUCCESS in 56m 56s\n- gate-grenade-dsvm http://logs.openstack.org/62/325462/1/check/gate-grenade-dsvm/eab59bd/ : SUCCESS in 24m 10s\n- gate-nova-releasenotes http://docs-draft.openstack.org/62/325462/1/check/gate-nova-releasenotes/2022cdb//releasenotes/build/html/ : SUCCESS in 3m 30s\n- gate-nova-tox-db-functional http://logs.openstack.org/62/325462/1/check/gate-nova-tox-db-functional/d2ab2ba/ : SUCCESS in 8m 58s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/62/325462/1/check/gate-grenade-dsvm-multinode/ce5832e/ : SUCCESS in 46m 59s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/310e221/ : SUCCESS in 23m 08s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-cells/8e13d31/ : SUCCESS in 27m 40s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/f7c6ad1/ : SUCCESS in 45m 50s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-multinode-full/f70c2fd/ : SUCCESS in 46m 25s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/62/325462/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/5166774/ : SUCCESS in 59m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bb013b3825f89facd71bb880369717f4a35fa65","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-06-04 06:09:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/62/325462/2/check/dsvm-tempest-kvm/fd15300 : SUCCESS in 1h 03m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"120110f83660e04a7c27f057234e3936dcbe9d8c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-06-04 09:09:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/62/325462/2/check/check-dsvm-tempest-vz7-exe-minimal/0f38092 : SUCCESS in 1h 09m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"3bb38c4477c53acc7155f71933e7bff1779191d5","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-06 04:45:09.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/62/325462/2/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/71b4d1617e3642cea9a50f4fe8f424fa/index.html : SUCCESS in 24m 41s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/62/325462/2/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6ccd061361044bd6b2403890492d9db1/index.html : SUCCESS in 22m 29s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/62/325462/2/check/real-db-upgrade_nova_mysql_user_001:th-mysql/b22e197f5ff547be9414d5d771728fb3/index.html : SUCCESS in 21m 24s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/62/325462/2/check/real-db-upgrade_nova_percona_user_001:th-percona/573737ddede64c018bf388565b850916/index.html : SUCCESS in 20m 02s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":2},{"id":"36a3971d44f7f1a4a6153acea2313df622e829ec","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2016-06-06 15:49:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a52bcf03db8acd8320bb96d69ce79c546eda1f32","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2016-06-07 15:13:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b9114fec7ac938beb2bdf18e4c7b639636c2dd53","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-07 17:04:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/325462/2/ : FAILURE in 33m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c113914ec28b86be5d81a145070c86346cd09c54","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 17:36:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/62/325462/2/check/gate-nova-docs/df277da//doc/build/html/ : SUCCESS in 4m 53s\n- gate-nova-pep8 http://logs.openstack.org/62/325462/2/check/gate-nova-pep8/40b969d/ : SUCCESS in 9m 27s\n- gate-nova-python27-db http://logs.openstack.org/62/325462/2/check/gate-nova-python27-db/1f77016/ : SUCCESS in 9m 48s\n- gate-nova-python34-db http://logs.openstack.org/62/325462/2/check/gate-nova-python34-db/b58d420/ : SUCCESS in 14m 03s\n- gate-nova-requirements http://logs.openstack.org/62/325462/2/check/gate-nova-requirements/58c2ece/ : SUCCESS in 2m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-full/501778a/ : SUCCESS in 1h 07m 26s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-postgres-full/552137e/ : SUCCESS in 45m 30s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-neutron-full/b6e7986/ : SUCCESS in 1h 22m 11s\n- gate-grenade-dsvm http://logs.openstack.org/62/325462/2/check/gate-grenade-dsvm/542a777/ : SUCCESS in 24m 19s\n- gate-nova-releasenotes http://docs-draft.openstack.org/62/325462/2/check/gate-nova-releasenotes/3e5d87c//releasenotes/build/html/ : SUCCESS in 4m 17s\n- gate-nova-tox-db-functional http://logs.openstack.org/62/325462/2/check/gate-nova-tox-db-functional/5519800/ : SUCCESS in 12m 14s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/62/325462/2/check/gate-grenade-dsvm-multinode/a9aea40/ : SUCCESS in 49m 25s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/a8935be/ : SUCCESS in 34m 07s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-cells/f55a22a/ : SUCCESS in 28m 42s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/e9855ee/ : SUCCESS in 38m 39s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-multinode-full/8c87647/ : SUCCESS in 48m 05s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/62/325462/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/46330ac/ : SUCCESS in 1h 00m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6580177673daaf8da528d9bc868eae40aadd70c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 17:36:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"6f992d39dddfa3375579e381e41330a66f472353","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-06-07 17:50:10.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm-recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/62/325462/2/check/check-ibm-tempest-dsvm-full/53e767c/ : SUCCESS in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbd55fb8a08ac0a3ddc35ec71b06a26ed931b3ad","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-07 19:25:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/62/325462/2/check/tempest-dsvm-full-nfv/4781881 : FAILURE in 1h 02m 12s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/62/325462/2/check/tempest-dsvm-intel-nfv/67c4cc9 : FAILURE in 51m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5dd29f52003749347774789d79825744f96f767a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-07 20:53:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/62/325462/2/gate/gate-nova-docs/e3f75f3//doc/build/html/ : SUCCESS in 10m 24s\n- gate-nova-pep8 http://logs.openstack.org/62/325462/2/gate/gate-nova-pep8/24d3e83/ : SUCCESS in 9m 05s\n- gate-nova-python27-db http://logs.openstack.org/62/325462/2/gate/gate-nova-python27-db/edd50b4/ : SUCCESS in 14m 12s\n- gate-nova-python34-db http://logs.openstack.org/62/325462/2/gate/gate-nova-python34-db/e213fcf/ : SUCCESS in 14m 27s\n- gate-nova-requirements http://logs.openstack.org/62/325462/2/gate/gate-nova-requirements/2ff7d8b/ : SUCCESS in 2m 00s\n- gate-tempest-dsvm-full http://logs.openstack.org/62/325462/2/gate/gate-tempest-dsvm-full/a76b595/ : SUCCESS in 37m 26s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/62/325462/2/gate/gate-tempest-dsvm-postgres-full/53a6a8a/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/62/325462/2/gate/gate-tempest-dsvm-neutron-full/a1e0aa7/ : SUCCESS in 55m 18s\n- gate-grenade-dsvm http://logs.openstack.org/62/325462/2/gate/gate-grenade-dsvm/f9e6108/ : SUCCESS in 26m 35s\n- gate-nova-releasenotes http://docs-draft.openstack.org/62/325462/2/gate/gate-nova-releasenotes/2ea7bd3//releasenotes/build/html/ : SUCCESS in 2m 25s\n- gate-nova-tox-db-functional http://logs.openstack.org/62/325462/2/gate/gate-nova-tox-db-functional/254f705/ : SUCCESS in 13m 09s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/62/325462/2/gate/gate-grenade-dsvm-multinode/21b3c64/ : SUCCESS in 50m 32s\n- gate-tempest-dsvm-cells http://logs.openstack.org/62/325462/2/gate/gate-tempest-dsvm-cells/c4075ed/ : SUCCESS in 31m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/62/325462/2/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/ea11172/ : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aa28428b8db8952365d8f0f9d08d42b89ef1ce17","date":"2016-06-07 20:53:17.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"597f079abeb6ae342ab518ab4445ba2502481601","revisions":{"db1c37222d5180e6746529111268a1b3c92e6906":{"kind":"REWORK","_number":1,"created":"2016-06-03 18:17:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/325462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/62/325462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/62/325462/1"}}},"commit":{"parents":[{"commit":"a224d5e8c0b45004c3f9109880eebee28f2954f3","subject":"Merge \"pci: Move PCI devices and PCI requests into migration context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a224d5e8c0b45004c3f9109880eebee28f2954f3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-03 18:17:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-03 18:17:11.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I79d5d82d32c157853de112e815089e71e16b39d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db1c37222d5180e6746529111268a1b3c92e6906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db1c37222d5180e6746529111268a1b3c92e6906"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a224d5e8c0b45004c3f9109880eebee28f2954f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"597f079abeb6ae342ab518ab4445ba2502481601":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-06-03 20:13:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/325462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/62/325462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/325462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/62/325462/2"}}},"commit":{"parents":[{"commit":"545d8d8666389f33601b0b003dec844004694919","subject":"Merge \"Fall back to flat config drive if not found in rbd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/545d8d8666389f33601b0b003dec844004694919"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-03 20:12:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2016-06-03 20:12:55.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I79d5d82d32c157853de112e815089e71e16b39d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/597f079abeb6ae342ab518ab4445ba2502481601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/597f079abeb6ae342ab518ab4445ba2502481601"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"545d8d8666389f33601b0b003dec844004694919","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
