)]}'
{"id":"openstack%2Fnova~327886","triplet_id":"openstack%2Fnova~master~I454ffb43450490503d988eff2654878670e5c35c","project":"openstack/nova","branch":"master","topic":"bug/1590929","hashtags":[],"change_id":"I454ffb43450490503d988eff2654878670e5c35c","subject":"Skip NFS and Ceph in live migration job test run","status":"MERGED","created":"2016-06-09 19:04:26.000000000","updated":"2016-06-15 03:47:07.000000000","submitted":"2016-06-14 19:48:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"757b442deac0ce2bf0978e60e2236f3968105813","_number":327886,"virtual_id_number":327886,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2016-06-14 20:45:48.000000000","post_submit":true,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2016-06-14 21:52:08.000000000","post_submit":true,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-14 19:48:45.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2016-06-15 03:47:07.000000000","post_submit":true,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2016-06-14 14:44:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-06-14 16:53:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2016-06-14 16:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":2,"date":"2016-06-14 14:12:10.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2016-06-14 16:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-09 23:13:12.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-10 01:16:07.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-14 14:12:10.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-06-14 14:44:23.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-14 16:53:19.000000000","updated_by":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"reviewer":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"state":"REVIEWER"},{"updated":"2016-06-14 16:55:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2016-06-14 19:48:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-06-14 20:45:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-14 21:52:08.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-06-15 03:47:07.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff3da6d2d006372856041367a96c76f3ad49a247","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-09 19:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"733a17e19730d76508072b9f0efe04740d6d68e4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-09 19:07:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4e5573620fe84bbd9b27922ac51fd60451070c47","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-09 19:08:08.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"7562fc390a2b67442b7bf74f7c647159445770d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 19:30:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/86/327886/2/check/gate-nova-docs/b9c7cf5//doc/build/html/ : SUCCESS in 5m 53s\n- gate-nova-pep8 http://logs.openstack.org/86/327886/2/check/gate-nova-pep8/4ae5063/ : SUCCESS in 5m 15s\n- gate-nova-python27-db http://logs.openstack.org/86/327886/2/check/gate-nova-python27-db/d93442c/ : FAILURE in 10m 58s\n- gate-nova-python34-db http://logs.openstack.org/86/327886/2/check/gate-nova-python34-db/3ecff15/ : SUCCESS in 14m 45s\n- gate-nova-tox-db-functional http://logs.openstack.org/86/327886/2/check/gate-nova-tox-db-functional/1703bed/ : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2a79d45354ab1d66ffb5a4d605268713fc496425","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 19:50:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/86/327886/2/experimental/gate-tempest-dsvm-multinode-live-migration/1474f08/ : FAILURE in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"231a38d6f6d6383ea583210117752a483234c621","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-09 20:15:15.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/327886/2/18452/index.html : SUCCESS in 01 h 05 m 29 s","accounts_in_message":[],"_revision_number":2},{"id":"3051029ead3dfd83d569af9b0e13b7abb328a4b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 20:34:48.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"15b4c62153fb2212ea23aeeb43ff9d6817f3d31e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-09 21:05:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"317275a87c7d55bb422363fca85b1d7797fa24a2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 21:24:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/86/327886/2/check/gate-nova-docs/312d4e6//doc/build/html/ : SUCCESS in 6m 14s\n- gate-nova-pep8 http://logs.openstack.org/86/327886/2/check/gate-nova-pep8/0c66723/ : SUCCESS in 6m 04s\n- gate-nova-python27-db http://logs.openstack.org/86/327886/2/check/gate-nova-python27-db/8f1c937/ : FAILURE in 7m 11s\n- gate-nova-python34-db http://logs.openstack.org/86/327886/2/check/gate-nova-python34-db/3ab94d0/ : SUCCESS in 12m 21s\n- gate-nova-tox-db-functional http://logs.openstack.org/86/327886/2/check/gate-nova-tox-db-functional/79e984c/ : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fa856b4d3e8b957ced2c134ed504932e9f8c8bc4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-09 22:00:43.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"c46f4c09128cdb90a2004050d2e445b4ef02d35b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 23:04:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/86/327886/2/experimental/gate-tempest-dsvm-multinode-live-migration/a7c62d7/ : FAILURE in 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2b9887b536a13bb5e9ceb93352911b1382dd955","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-09 23:13:12.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/327886/2/ : FAILURE in 2h 50m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d39160aa5cd210f97813da8606ea267b0db1e26e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-09 23:15:39.000000000","message":"Patch Set 2:\n\nBuild failed (experimental-tripleo).\n\n- gate-tripleo-ci-centos-7-nonha http://logs.openstack.org/86/327886/2/experimental-tripleo/gate-tripleo-ci-centos-7-nonha/cf93bb5/ : FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0ff6fedcfe08b3341a5f6c328579d7a1e50f9d49","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-10 00:18:46.000000000","message":"Patch Set 2:\n\nceph ephemeral setup also fails, initctl isn\u0027t on the system. Reported a bug:\n\nhttps://bugs.launchpad.net/nova/+bug/1591001","accounts_in_message":[],"_revision_number":2},{"id":"1a9de0eb230edaf540927bf1e96669126062f838","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-10 01:01:49.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/327886/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"30350a403f8564447b532e23baa84e80968f52be","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-10 01:16:07.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/327886/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"427a368cc3a5141719472fae688e0fcb33426ca4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-14 14:11:23.000000000","message":"Patch Set 3: Published edit on patch set 2","accounts_in_message":[],"_revision_number":3},{"id":"6b4b1beef16b8eff33d368d414511a75a7bae51e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-06-14 14:12:10.000000000","message":"Patch Set 3: Code-Review+2\n\nSeems worth a try.","accounts_in_message":[],"_revision_number":3},{"id":"3ec19f6bc7bf481bbae810a44cd1736806cbbef0","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-06-14 14:12:20.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"03b64cae0a34503dc0d95a68b1624a333cc6b993","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-14 14:13:02.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"934897fd9601d0fc35711206860831fecd7c9e88","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-14 14:13:12.000000000","message":"Patch Set 3: Workflow-1\n\nLet\u0027s make sure the live migration job passes first.","accounts_in_message":[],"_revision_number":3},{"id":"df93bf372fff938fb9cbc9231f037af1b8e6a0d3","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-06-14 14:14:09.000000000","message":"Patch Set 3:\n\n+1 on getting the pass first","accounts_in_message":[],"_revision_number":3},{"id":"19ec950be2d150265c55ad8010a95f3e52bb0a5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 14:39:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/86/327886/3/check/gate-nova-docs/4dd0c64//doc/build/html/ : SUCCESS in 4m 15s\n- gate-nova-pep8 http://logs.openstack.org/86/327886/3/check/gate-nova-pep8/d45883f/ : SUCCESS in 4m 57s\n- gate-nova-python27-db http://logs.openstack.org/86/327886/3/check/gate-nova-python27-db/ea48992/ : SUCCESS in 8m 07s\n- gate-nova-python34-db http://logs.openstack.org/86/327886/3/check/gate-nova-python34-db/c5dc857/ : SUCCESS in 11m 44s\n- gate-nova-tox-db-functional http://logs.openstack.org/86/327886/3/check/gate-nova-tox-db-functional/5b82a5d/ : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":3},{"id":"03ea605b5e2958432080c576326179d4c9f1bb4b","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2016-06-14 14:44:23.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/86/327886/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/7712c953cc53456b8e772f03cd2e5301/index.html : SUCCESS in 17m 49s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/86/327886/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/87e5eba799f94c02bf937f4bf1e9454e/index.html : SUCCESS in 17m 53s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/86/327886/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7850e042d3ae4500a7b1f9be8c2a1dcc/index.html : SUCCESS in 16m 17s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/86/327886/3/check/real-db-upgrade_nova_percona_user_001:th-percona/58e110fa637845ce90ff6b01d05a6291/index.html : SUCCESS in 20m 38s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":3},{"id":"ab1b3f87a79b2ef706acbe4c7d7b9a0e85d53da6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 14:56:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/86/327886/3/experimental/gate-tempest-dsvm-multinode-live-migration/475fbc3/ : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37c53a99a3e867a2979998f75c0b80425f3eef9c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-14 16:32:38.000000000","message":"Patch Set 3: -Workflow\n\nThe live migration job passed:\n\nhttp://logs.openstack.org/86/327886/3/experimental/gate-tempest-dsvm-multinode-live-migration/475fbc3/console.html","accounts_in_message":[],"_revision_number":3},{"id":"5ccb36daa70fc4ef616573661362d3e4b2709d62","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-06-14 16:36:39.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/327886/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"07a699c1d21fe141449d5b53592772dd9259d4f9","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2016-06-14 16:53:19.000000000","message":"Patch Set 3: Code-Review+1\n\nlet\u0027s try to gather stats for simple scenario without shared storages","accounts_in_message":[],"_revision_number":3},{"id":"0b401893e70ed579c272f828a0ada58a55a12255","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-06-14 16:55:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"29d8c6ea90f8b7636d331410ae69524094cb6b40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 16:55:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"266ca78e7c19dab38210343d419a848196e568ed","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-06-14 16:55:44.000000000","message":"Patch Set 3:\n\nCome to think of it, if we\u0027re just running the basic config, I think we already test that in the non-voting multinode job, so getting this going probably doesn\u0027t buy us much.","accounts_in_message":[],"_revision_number":3},{"id":"9d68291346ace0d38a2657517022dac708ddb947","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-06-14 18:23:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/327886/3/ : SUCCESS in 2h 24m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ebceccd0d64e0825358abb3ae7f93004a09fd096","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-14 19:48:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/86/327886/3/gate/gate-nova-docs/ec5c355//doc/build/html/ : SUCCESS in 4m 27s\n- gate-nova-pep8 http://logs.openstack.org/86/327886/3/gate/gate-nova-pep8/9f54c16/ : SUCCESS in 5m 51s\n- gate-nova-python27-db http://logs.openstack.org/86/327886/3/gate/gate-nova-python27-db/660c321/ : SUCCESS in 10m 42s\n- gate-nova-python34-db http://logs.openstack.org/86/327886/3/gate/gate-nova-python34-db/d32b93a/ : SUCCESS in 11m 56s\n- gate-nova-tox-db-functional http://logs.openstack.org/86/327886/3/gate/gate-nova-tox-db-functional/15165d2/ : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0a2c91aeecf02b7cbd18a7dcacd6344bd6d1f239","date":"2016-06-14 19:48:47.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"98c50e641d9513a12d55adfec5972041b6b602df","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-06-14 20:45:48.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-86-327886-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"2f2f2729acbef1836b26b969f9b89d9c2f4150c3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-06-14 21:52:08.000000000","message":"Patch Set 3:\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/86/327886/3/19322/index.html : SUCCESS in 01 h 02 m 12 s","accounts_in_message":[],"_revision_number":3},{"id":"9e10cae58943809b8c704ec3a0626abe65711538","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-15 03:07:31.000000000","message":"Patch Set 3:\n\nBuild failed (experimental-tripleo).\n\n- gate-tripleo-ci-centos-7-nonha http://logs.openstack.org/86/327886/3/experimental-tripleo/gate-tripleo-ci-centos-7-nonha/af8e689/ : FAILURE in 17m 08s","accounts_in_message":[],"_revision_number":3},{"id":"56f9eeb696f1791c15efe2a0cd2b68972cbdb771","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-06-15 03:47:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/86/327886/3/check/tempest-dsvm-full-nfv/2e8a24d : SUCCESS in 1h 28m 43s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/86/327886/3/check/tempest-dsvm-intel-nfv/90a522f : FAILURE in 32m 58s\n- tempest-dsvm-ovsdpdk-nfv http://intel-openstack-ci-logs.ovh/86/327886/3/check/tempest-dsvm-ovsdpdk-nfv/d6ae66c : SUCCESS in 1h 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"69760332f9837fbf6bd89e56d4d77a9b3252fc11","revisions":{"4be096a8428147ea203148360bbd31b1432f451e":{"kind":"REWORK","_number":1,"created":"2016-06-09 19:04:26.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/86/327886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/327886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/327886/1"}}},"commit":{"parents":[{"commit":"caa30a69350366954769201ded46dbae6a2b68e0","subject":"Merge \"Add ResourceProvider.save()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/caa30a69350366954769201ded46dbae6a2b68e0"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-09 19:02:37.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-09 19:02:37.000000000","tz":-240},"subject":"Skip NFS in live migration job test run","message":"Skip NFS in live migration job test run\n\nThe NFS setup is currently broken on the Ubuntu 16.04 nodes\nso until that\u0027s sorted out let\u0027s skip it.\n\nDepends on two changes, one to fix a block migration +\ntunneling bug in Nova and another to fix the Ceph setup.\n\nChange-Id: I454ffb43450490503d988eff2654878670e5c35c\nDepends-On: Id6e49f298133c53d21386ea619c83e413ef3117a\nDepends-On: I3255f9421119aafb7407ba7443400bd3ace0d827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4be096a8428147ea203148360bbd31b1432f451e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4be096a8428147ea203148360bbd31b1432f451e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"caa30a69350366954769201ded46dbae6a2b68e0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c192b19468df5745efa705b72f5543dc2aa5688e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2016-06-09 19:07:23.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/86/327886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/327886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/327886/2"}}},"commit":{"parents":[{"commit":"caa30a69350366954769201ded46dbae6a2b68e0","subject":"Merge \"Add ResourceProvider.save()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/caa30a69350366954769201ded46dbae6a2b68e0"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-09 19:02:37.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-09 19:06:56.000000000","tz":-240},"subject":"Skip NFS in live migration job test run","message":"Skip NFS in live migration job test run\n\nThe NFS setup is currently broken on the Ubuntu 16.04 nodes\nso until that\u0027s sorted out let\u0027s skip it.\n\nDepends on two changes, one to fix a block migration +\ntunneling bug in Nova and another to fix the Ceph setup.\n\nRelated-Bug: #1590929\n\nChange-Id: I454ffb43450490503d988eff2654878670e5c35c\nDepends-On: Id6e49f298133c53d21386ea619c83e413ef3117a\nDepends-On: I3255f9421119aafb7407ba7443400bd3ace0d827\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c192b19468df5745efa705b72f5543dc2aa5688e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c192b19468df5745efa705b72f5543dc2aa5688e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"caa30a69350366954769201ded46dbae6a2b68e0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"69760332f9837fbf6bd89e56d4d77a9b3252fc11":{"kind":"REWORK","_number":3,"created":"2016-06-14 14:11:23.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/86/327886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/327886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/327886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/327886/3"}}},"commit":{"parents":[{"commit":"caa30a69350366954769201ded46dbae6a2b68e0","subject":"Merge \"Add ResourceProvider.save()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/caa30a69350366954769201ded46dbae6a2b68e0"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-09 19:02:37.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-06-14 14:11:20.000000000","tz":0},"subject":"Skip NFS and Ceph in live migration job test run","message":"Skip NFS and Ceph in live migration job test run\n\nThe NFS and Ceph setup is currently broken on the Ubuntu 16.04 nodes\nso until that\u0027s sorted out let\u0027s skip them.\n\nRelated-Bug: #1590929\n\nChange-Id: I454ffb43450490503d988eff2654878670e5c35c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69760332f9837fbf6bd89e56d4d77a9b3252fc11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69760332f9837fbf6bd89e56d4d77a9b3252fc11"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"caa30a69350366954769201ded46dbae6a2b68e0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
