)]}'
{"id":"openstack%2Fnova~338553","triplet_id":"openstack%2Fnova~master~Idc049b455a7df483556cfba0c34d5fcfb3deb3ec","project":"openstack/nova","branch":"master","topic":"bp/centralize-config-options-ocata","hashtags":[],"change_id":"Idc049b455a7df483556cfba0c34d5fcfb3deb3ec","subject":"Remove deprecated manager classes in service.py","status":"ABANDONED","created":"2016-07-06 21:04:27.000000000","updated":"2016-11-14 18:15:13.000000000","total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a83ea511b9305076c5ac7dd8d72568ead2816e96","_number":338553,"virtual_id_number":338553,"owner":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2016-11-09 20:53:28.000000000","_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"date":"2016-11-09 21:36:34.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-11-09 20:57:58.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2016-11-10 22:54:51.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2016-11-09 21:22:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"date":"2016-11-09 21:04:21.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"date":"2016-11-14 16:19:17.000000000","_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},{"date":"2016-11-09 21:57:28.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":1,"date":"2016-11-09 22:35:10.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2016-11-09 21:55:56.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"date":"2016-11-09 21:43:12.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":1,"date":"2016-11-10 07:05:52.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-27 16:20:29.000000000","updated_by":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"reviewer":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"state":"REVIEWER"},{"updated":"2016-09-28 08:59:05.000000000","updated_by":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"reviewer":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"state":"REVIEWER"},{"updated":"2016-10-06 14:10:46.000000000","updated_by":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"reviewer":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"state":"REVIEWER"},{"updated":"2016-10-10 16:46:47.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-10-12 19:13:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-13 00:00:41.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2016-10-13 16:30:25.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-10-14 09:13:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-10-18 11:37:58.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2016-11-09 20:53:28.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 20:57:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-09 21:04:21.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 21:22:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 21:36:34.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 21:43:12.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 21:55:56.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 21:57:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-09 22:35:10.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-10 07:05:52.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-10 22:54:51.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-14 16:19:17.000000000","updated_by":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"reviewer":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"state":"REVIEWER"}],"messages":[{"id":"42a84ce9aab7f9777b24a8e115944d283489dca7","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-06 21:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0077edb42b02b3eef429a0f6c236d73738d8acf6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-06 22:36:46.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/1/ : FAILURE in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f2643437f6d94d0170ccfa97afbb7e0ae568ee35","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-06 22:41:50.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/338553/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"377a7864c628a121dab0f96d4fc5a4057e0f1232","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-06 22:52:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/1/check/dsvm-tempest-xen/e37eb2d : FAILURE in 1h 47m 08s","accounts_in_message":[],"_revision_number":1},{"id":"675af0628fe351ba2eaa3e557db82f3f8f1abdb1","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-06 22:55:19.000000000","message":"Patch Set 1:\n\nxen: recheck","accounts_in_message":[],"_revision_number":1},{"id":"ab297ecba48ebdc9cc1462f97a98effbd16b4b48","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-06 22:56:57.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/1/check/check-ibm-tempest-dsvm-full/c43ea57/ : FAILURE in 1h 52m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e174ce9ef54c91c3a8738cf6e0bb180637201560","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-06 23:23:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/1/check/gate-nova-docs/35a8ba3//doc/build/html/ : SUCCESS in 5m 42s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/1/check/gate-nova-pep8/73ea7fc/ : SUCCESS in 7m 34s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/1/check/gate-nova-python27-db/2acb6d0/ : SUCCESS in 15m 06s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/1/check/gate-nova-python34-db/10194ff/ : SUCCESS in 8m 31s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/1/check/gate-nova-python35-db-nv/25fb2a6/ : SUCCESS in 15m 15s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-full/42df89a/ : FAILURE in 1h 50m 14s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-postgres-full/b84a2c6/ : FAILURE in 1h 51m 23s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-neutron-full/a9cf278/ : FAILURE in 2h 17m 19s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/1/check/gate-grenade-dsvm/d5ff26c/ : FAILURE in 51m 24s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/1/check/gate-nova-releasenotes/8517603//releasenotes/build/html/ : SUCCESS in 4m 03s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/1/check/gate-nova-tox-db-functional/868bb15/ : FAILURE in 13m 55s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/1/check/gate-grenade-dsvm-multinode/a62d898/ : FAILURE in 51m 24s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/5c9ef62/ : FAILURE in 18m 21s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-cells/c720201/ : FAILURE in 1h 29m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/94c41ca/ : FAILURE in 1h 47m 37s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-multinode-full/a08d3ab/ : FAILURE in 1h 55m 52s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/5801d92/ : FAILURE in 2h 10m 23s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-multinode-live-migration/2e38310/ : FAILURE in 35m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efb33ad195754de944345a1311037409ca7272d9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-06 23:48:24.000000000","message":"Patch Set 1: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/1/23592/index.html : FAILURE in 02 h 41 m 02 s","accounts_in_message":[],"_revision_number":1},{"id":"72f77c45e028ae7b319a541d02aab33d00c43b76","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-07 00:33:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-full-nfv/dbee819 : FAILURE in 2h 48m 44s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-intel-nfv/1d705ab : FAILURE in 51m 33s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-ovsdpdk-nfv-networking/2cb73a7 : FAILURE in 1h 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"820151e4a357fceb6193045959b4d739b1bb9fa3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-07 00:49:09.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/1/check/dsvm-tempest-xen/fcb37e9 : FAILURE in 1h 46m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3ebc58bfe49d35e3c7edd968030a6a4e82e9d4a5","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-07 04:16:06.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3143fd768640841396ca46688d9d73f6e35e719e","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-07 20:46:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b888bfa9e6a90090afc93a68512ab50c683e45e5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-07 22:43:06.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/1/check/check-ibm-tempest-dsvm-full/e36ac63/ : FAILURE in 1h 50m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25b0297362af45d8c8ea52b13b61c4a7bf6890de","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-07-07 23:04:45.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/1/check/check-dsvm-tempest-vz7-exe-minimal/9513085 : FAILURE in 2h 15m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"93b6efc4a3524d9d18167dde600ad9dd7e2e38da","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-08 00:12:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/1/check/gate-nova-docs/3cff5d3//doc/build/html/ : SUCCESS in 6m 12s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/1/check/gate-nova-pep8/2e1d2a3/ : SUCCESS in 7m 12s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/1/check/gate-nova-python27-db/0c47ea3/ : SUCCESS in 15m 09s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/1/check/gate-nova-python34-db/f6e1b77/ : SUCCESS in 9m 24s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/1/check/gate-nova-python35-db-nv/1a036e5/ : SUCCESS in 13m 27s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-full/ae42910/ : FAILURE in 1h 49m 43s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-postgres-full/9f94530/ : FAILURE in 2h 03m 34s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-neutron-full/fb60bc8/ : FAILURE in 2h 47m 20s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/1/check/gate-grenade-dsvm/a7592ea/ : FAILURE in 38m 49s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/1/check/gate-nova-releasenotes/7773f09//releasenotes/build/html/ : SUCCESS in 4m 55s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/1/check/gate-nova-tox-db-functional/367c20b/ : FAILURE in 13m 30s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/1/check/gate-grenade-dsvm-multinode/0f6b275/ : FAILURE in 1h 00m 27s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/5ced0dd/ : FAILURE in 28m 28s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-cells/99b618a/ : FAILURE in 1h 40m 14s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/dfe913c/ : FAILURE in 2h 01m 55s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-multinode-full/b8d3e0a/ : FAILURE in 1h 56m 15s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/ff0b98f/ : FAILURE in 2h 47m 34s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/1/check/gate-tempest-dsvm-multinode-live-migration/a6bb475/ : FAILURE in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20bb3e4eedc8dc1d8ac03eb0cc638543137ff9a4","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-08 00:34:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/1/ : FAILURE in 1h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"105fb31c566e5cb0a2087c6e837247b044ad1dfa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-08 02:39:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-full-nfv/cc4485e : FAILURE in 2h 49m 11s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-intel-nfv/f82b179 : FAILURE in 53m 57s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/1/check/tempest-dsvm-ovsdpdk-nfv-networking/4657290 : FAILURE in 1h 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7151c21b8926fc89bbde10e289127b0fd0cf3b2d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-07-09 21:28:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* dsvm-ext-nova http://208.91.1.172/logs/dsvm-ext-nova/338553/1/1178/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"771367ebd3aa31460a52f25167a3291602a507bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-07-09 23:38:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* dsvm-ext-nova http://208.91.1.172/logs/dsvm-ext-nova/338553/1/1182/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fde5a69636e15c629eb4951e58bdd2cc54b2331f","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-11 15:20:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cb9054222db59285fd33885e66ed6c2160fe011","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-11 15:36:51.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/338553/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"52e7a81e67ac5f487077f1f5e9c1045af9262c9e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-11 16:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/2/ : FAILURE in 1h 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13454545f0e3592f5c1fad581414521285230b63","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-11 17:11:08.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/2/check/check-ibm-tempest-dsvm-full/cd69ebe/ : FAILURE in 1h 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c29ff1939f9c176bfd07baa6f83e7d0924c2968","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-11 17:17:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/2/check/dsvm-tempest-xen/6682846 : FAILURE in 1h 47m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a187e3a6f27f8c35b6fe936a73e00a1d4fa36d79","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-11 17:43:10.000000000","message":"Patch Set 2: Verified-1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/2/24146/index.html : FAILURE in 02 h 17 m 56 s","accounts_in_message":[],"_revision_number":2},{"id":"bf7ebec313b5bc5d881e57f838ccb1f3ec42f48c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-11 18:45:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/2/check/gate-nova-docs/2eff1c6//doc/build/html/ : SUCCESS in 3m 21s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/2/check/gate-nova-pep8/e83a4d1/ : SUCCESS in 4m 36s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/2/check/gate-nova-python27-db/23eda02/ : SUCCESS in 10m 38s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/2/check/gate-nova-python34-db/b1f0c8b/ : SUCCESS in 13m 12s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/2/check/gate-nova-python35-db-nv/3177d3a/ : SUCCESS in 14m 10s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-full/5305ee0/ : FAILURE in 1h 48m 59s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-postgres-full/d41eca3/ : FAILURE in 1h 49m 45s (non-voting)\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-neutron-full/2951d2c/ : FAILURE in 2h 22m 12s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/2/check/gate-grenade-dsvm/1f242cc/ : FAILURE in 55m 34s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/2/check/gate-nova-releasenotes/a81a044//releasenotes/build/html/ : SUCCESS in 5m 30s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/2/check/gate-nova-tox-db-functional/5320a35/ : SUCCESS in 14m 56s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/2/check/gate-grenade-dsvm-multinode/4384071/ : FAILURE in 1h 09m 43s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/7d83b51/ : FAILURE in 29m 24s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-cells/80a9513/ : FAILURE in 1h 33m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph/d7131f2/ : FAILURE in 1h 49m 37s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-multinode-full/c1ab138/ : FAILURE in 2h 01m 32s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/d54e151/ : FAILURE in 2h 27m 18s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/2/check/gate-tempest-dsvm-multinode-live-migration/5399aaf/ : FAILURE in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a54936be35761a45789592fb0d31b66130abe4a3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-07-11 18:47:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/2/check/dsvm-tempest-kvm/007483a : FAILURE in 1h 41m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"7f201b265d345492c775dd81a941702672cc8ade","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-11 18:48:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/2/check/tempest-dsvm-full-nfv/de43d7c : FAILURE in 2h 48m 40s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/2/check/tempest-dsvm-intel-nfv/f5ecd37 : FAILURE in 58m 54s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/2/check/tempest-dsvm-ovsdpdk-nfv-networking/5690270 : FAILURE in 1h 19m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7f995c96996d7e7ca6413fbc752367ce421518af","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-12 18:42:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6eefa59c0e4dca77232040c93a8e38bda6472266","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-12 19:50:13.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/3/check/check-ibm-tempest-dsvm-full/658b3bb/ : SUCCESS in 1h 06m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"07c694cab222c8def46c54d6d431423ebb7a76d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-12 20:17:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/3/check/gate-nova-docs/8af2101//doc/build/html/ : SUCCESS in 5m 17s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/3/check/gate-nova-pep8/b1f4b91/ : SUCCESS in 7m 15s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/3/check/gate-nova-python27-db/a399ca0/ : SUCCESS in 13m 02s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/3/check/gate-nova-python34-db/005323e/ : SUCCESS in 9m 11s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/3/check/gate-nova-python35-db-nv/20f7c21/ : SUCCESS in 13m 55s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full/e29e4d2/ : SUCCESS in 34m 27s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-postgres-full/62d7e6d/ : FAILURE in 51m 13s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-full/eee13fb/ : SUCCESS in 1h 17m 11s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm/1efb1db/ : SUCCESS in 44m 46s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/3/check/gate-nova-releasenotes/cf0b169//releasenotes/build/html/ : SUCCESS in 4m 59s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/3/check/gate-nova-tox-db-functional/8035e57/ : SUCCESS in 16m 16s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm-multinode/b89f7b0/ : SUCCESS in 48m 03s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/fc7a777/ : SUCCESS in 37m 40s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-cells/bf17964/ : FAILURE in 36m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/3c8576f/ : SUCCESS in 59m 02s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-full/02523a4/ : FAILURE in 56m 48s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/1c137ac/ : SUCCESS in 1h 00m 33s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-live-migration/2a4f466/ : SUCCESS in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f4ade96e0d63b87fb7b9ca7bfa151e2a1a4a1bc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-12 20:21:50.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/3/24421/index.html : SUCCESS in 01 h 03 m 51 s","accounts_in_message":[],"_revision_number":3},{"id":"bd4c06aad157aab19232a610f14cdbcaa765a659","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-12 20:31:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3d9d208c39d9459a5a3ab8fdedd6c75b5bfbc120","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-07-12 20:40:16.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells: unrecognized error\n- gate-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1355573 https://bugs.launchpad.net/bugs/1349617\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"5065d300d70c6349fbb1902569e21ff7cdb65d08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-12 21:11:05.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/3/check/check-ibm-tempest-dsvm-full/66b8a9b/ : FAILURE in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b53c36c6b5ffbba03e7be31ab803f0297b3a2838","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-12 21:43:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/3/check/gate-nova-docs/72d4b90//doc/build/html/ : SUCCESS in 2m 53s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/3/check/gate-nova-pep8/97afa88/ : SUCCESS in 5m 14s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/3/check/gate-nova-python27-db/aefa473/ : SUCCESS in 11m 00s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/3/check/gate-nova-python34-db/d57e903/ : SUCCESS in 9m 17s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/3/check/gate-nova-python35-db-nv/b3e4c02/ : SUCCESS in 12m 33s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full/f1ca18f/ : SUCCESS in 45m 49s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-postgres-full/0c3cf15/ : FAILURE in 55m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-full/90e5978/ : SUCCESS in 1h 09m 58s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm/efe7eb4/ : SUCCESS in 39m 37s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/3/check/gate-nova-releasenotes/d92aa12//releasenotes/build/html/ : SUCCESS in 2m 51s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/3/check/gate-nova-tox-db-functional/74f09e9/ : SUCCESS in 15m 17s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm-multinode/7b0da4a/ : SUCCESS in 48m 51s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/6772e0e/ : SUCCESS in 36m 05s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-cells/3e298f7/ : SUCCESS in 53m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/5f2ee22/ : SUCCESS in 58m 25s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-full/6831047/ : FAILURE in 58m 21s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/affebe2/ : SUCCESS in 1h 02m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-live-migration/231f112/ : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1fc7ff0cca8578a7e3080f5df37866fc5c354ca7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-07-12 21:43:28.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1355573 https://bugs.launchpad.net/bugs/1349617\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"500cc2a6a4cdc8d1c1e57535d7f69d6758d38243","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-12 22:15:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"931fe00d99442eb8d8aea135d5c711ae62964b3e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-12 22:34:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/3/check/dsvm-tempest-xen/32d2f38 : SUCCESS in 1h 25m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cf350b896b30e7a4773b5733beba24d8dcaf7511","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-12 23:17:07.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/3/check/check-ibm-tempest-dsvm-full/fd94f67/ : SUCCESS in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eba8366850a9b5bbea2400105a33fdecaf5caa9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-12 23:19:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/3/check/gate-nova-docs/da39d5b//doc/build/html/ : SUCCESS in 4m 35s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/3/check/gate-nova-pep8/3be9ca6/ : SUCCESS in 6m 41s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/3/check/gate-nova-python27-db/2e36f3a/ : SUCCESS in 13m 04s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/3/check/gate-nova-python34-db/5845e36/ : SUCCESS in 9m 20s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/3/check/gate-nova-python35-db-nv/b64b3e4/ : SUCCESS in 12m 53s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full/7098b2e/ : SUCCESS in 49m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-postgres-full/d75719c/ : FAILURE in 55m 37s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-full/9803597/ : SUCCESS in 49m 28s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm/55c8632/ : SUCCESS in 42m 54s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/3/check/gate-nova-releasenotes/e4d9eef//releasenotes/build/html/ : SUCCESS in 4m 02s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/3/check/gate-nova-tox-db-functional/9989945/ : SUCCESS in 17m 59s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/3/check/gate-grenade-dsvm-multinode/54f4006/ : SUCCESS in 53m 33s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/59a1673/ : SUCCESS in 42m 52s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-cells/7cd9d78/ : SUCCESS in 37m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph/f33afa3/ : SUCCESS in 50m 52s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-full/a2c91f6/ : FAILURE in 58m 32s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/91101fd/ : SUCCESS in 57m 16s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/3/check/gate-tempest-dsvm-multinode-live-migration/96e5b08/ : SUCCESS in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f3d5bf8825e9752e53637279d4bc6fe7527d88b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-12 23:23:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/3/check/tempest-dsvm-full-nfv/92e767f : SUCCESS in 1h 31m 07s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/3/check/tempest-dsvm-intel-nfv/d8f0ea5 : SUCCESS in 34m 49s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/3/check/tempest-dsvm-ovsdpdk-nfv-networking/547e41e : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":3},{"id":"a1318b7e7b150267e092049b04f72deb299aa519","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-13 00:18:21.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/338553/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"7468b77ca655735a85896392156030f10d9648f7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 02:00:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/3/ : SUCCESS in 2h 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef99cdf511232df4a3000e314efb84f379d68869","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-07-13 02:33:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild Successful \n\n* dsvm-ext-nova http://208.91.1.172/logs/dsvm-ext-nova/338553/3/1314/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6357ea55bae3353f506cf8914512808a19343787","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 07:09:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/3/check/dsvm-tempest-kvm/1ed2833 : FAILURE in 18m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"f5562c872ee723e1f051bf3c28f85616283cbaa3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 13:26:04.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/3/check/check-dsvm-tempest-vz7-exe-minimal/e3ec2d6 : FAILURE in 2h 06m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"9d9cf9857ceeca31a64dc02959245e079250a1df","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 14:28:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb01059699a2a697af5de5e49461c5d12a353d16","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 15:07:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"552060018ea90235bff85e57629d60ba7f27ca23","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-13 15:31:01.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/338553/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"3c8c6dfa66384c8c351c15cfb33325f121ad51e6","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 16:01:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/4/ : FAILURE in 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ffd592ddef5507fff475d7d599edb31f0057264","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-13 16:12:35.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/4/check/check-ibm-tempest-dsvm-full/7a5186b/ : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b27bc197b35fc0bb013d39e89ababac15a7b20a1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-13 16:24:40.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/4/24530/index.html : SUCCESS in 01 h 14 m 06 s","accounts_in_message":[],"_revision_number":4},{"id":"6ead6c371caa2093d79ea69ce1077f6a1035544c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-13 16:33:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/4/check/dsvm-tempest-xen/af829eb : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":4},{"id":"51ec9f736cfc667935a30e110ac3787b15aadb5a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 16:33:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/4/check/gate-nova-docs/8975750//doc/build/html/ : SUCCESS in 5m 25s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/4/check/gate-nova-pep8/6ec5cf8/ : SUCCESS in 10m 15s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/4/check/gate-nova-python27-db/2f1d48e/ : SUCCESS in 10m 29s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/4/check/gate-nova-python34-db/06ba042/ : SUCCESS in 11m 17s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/4/check/gate-nova-python35-db-nv/755ed93/ : SUCCESS in 12m 15s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-full/4978feb/ : SUCCESS in 45m 23s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-postgres-full/e555cf7/ : FAILURE in 55m 42s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-neutron-full/d399338/ : SUCCESS in 1h 09m 58s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/4/check/gate-grenade-dsvm/947a9ff/ : SUCCESS in 44m 35s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/4/check/gate-nova-releasenotes/1905f5d//releasenotes/build/html/ : SUCCESS in 5m 11s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/4/check/gate-nova-tox-db-functional/197c6cb/ : SUCCESS in 14m 15s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/4/check/gate-grenade-dsvm-multinode/4e423ec/ : SUCCESS in 46m 08s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/8a9184a/ : SUCCESS in 33m 21s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-cells/da31dd5/ : SUCCESS in 38m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph/0daa3f8/ : SUCCESS in 48m 06s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-multinode-full/682744d/ : FAILURE in 1h 02m 11s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/83c3397/ : SUCCESS in 1h 03m 12s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/4/check/gate-tempest-dsvm-multinode-live-migration/3829898/ : SUCCESS in 31m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4da5b4f8dda2b976e04f92afdb5b95fe8e9106d5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 16:35:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/4/check/tempest-dsvm-full-nfv/156306a : FAILURE in 1h 24m 45s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/4/check/tempest-dsvm-intel-nfv/38e67a1 : SUCCESS in 40m 13s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/4/check/tempest-dsvm-ovsdpdk-nfv-networking/9c65574 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ecf2dd736f8cc00c2fee3547514cbe38144be3b6","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 17:52:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9c7b2c7829e78d9ac247bf900cbdb3869530ec42","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 18:04:41.000000000","message":"Patch Set 5: Published edit on patch set 4","accounts_in_message":[],"_revision_number":5},{"id":"ccf230418d1970841d1929d53e9a8f2f567abade","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-13 19:12:26.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/5/check/check-ibm-tempest-dsvm-full/74f2c43/ : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc942e4120f88a940a85a209f1386b9f69c29277","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-13 19:13:47.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/338553/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"15920e72898250750480b43bc08f527f0ffd4b40","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-13 19:19:08.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/5/24552/index.html : SUCCESS in 01 h 05 m 23 s","accounts_in_message":[],"_revision_number":5},{"id":"90884804491460108032bbd0da4dc599d8250b40","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-13 19:36:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/5/check/dsvm-tempest-xen/0dd052c : SUCCESS in 1h 22m 56s","accounts_in_message":[],"_revision_number":5},{"id":"125ebc5c07145e5807a8f19a566dbe4857a02351","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 19:53:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/5/check/gate-nova-docs/8e1d665//doc/build/html/ : SUCCESS in 4m 31s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/5/check/gate-nova-pep8/e083653/ : SUCCESS in 8m 15s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/5/check/gate-nova-python27-db/a2f1ee0/ : SUCCESS in 15m 16s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/5/check/gate-nova-python34-db/a25bc6b/ : SUCCESS in 11m 46s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/5/check/gate-nova-python35-db-nv/81ecec8/ : SUCCESS in 10m 57s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-full/cf537b7/ : SUCCESS in 46m 08s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-postgres-full/f0905cf/ : SUCCESS in 39m 47s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-neutron-full/da64325/ : SUCCESS in 1h 03m 06s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/5/check/gate-grenade-dsvm/11a5a8e/ : SUCCESS in 29m 01s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/5/check/gate-nova-releasenotes/31f6013//releasenotes/build/html/ : SUCCESS in 6m 07s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/5/check/gate-nova-tox-db-functional/a8ea037/ : SUCCESS in 16m 50s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/5/check/gate-grenade-dsvm-multinode/b173653/ : SUCCESS in 46m 29s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/44813bf/ : SUCCESS in 35m 48s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-cells/544129f/ : SUCCESS in 44m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph/f9ba5f6/ : SUCCESS in 44m 35s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-multinode-full/d05e651/ : FAILURE in 56m 26s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/53dcd7d/ : SUCCESS in 56m 24s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/5/check/gate-tempest-dsvm-multinode-live-migration/0495607/ : SUCCESS in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4e71ddf625f82d7da72f57b58c20252acabe7870","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 20:28:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/5/ : FAILURE in 2h 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5eb9ff7f73d227e376ebc551afd0de0c8ed99cee","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 21:07:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/5/check/tempest-dsvm-full-nfv/4c3727a : FAILURE in 2h 47m 37s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/5/check/tempest-dsvm-intel-nfv/524d44c : SUCCESS in 41m 22s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/5/check/tempest-dsvm-ovsdpdk-nfv-networking/631fef9 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":5},{"id":"e57716afb401c79649b686d794f99d51b29722da","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 21:12:38.000000000","message":"Patch Set 6: Published edit on patch set 5","accounts_in_message":[],"_revision_number":6},{"id":"95f10e50899d9f399b4e9ca44c8fe70656269d12","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-13 21:13:13.000000000","message":"Patch Set 7: Published edit on patch set 6","accounts_in_message":[],"_revision_number":7},{"id":"bda523992ef724d87f181b68af19fa9a92cd744e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-13 21:57:57.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/338553/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"c665704be8ee1a393adcb13029d726b7c45131ff","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-13 21:59:12.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/338553/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a96fd4c5a9cbf02988b559180fc163bfeeab82b5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-13 22:26:24.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on Citrix Xenserver. For rechecking only on the Citrix Xenserver CI,add a review comment with \"xenserver: recheck\". Contact info: openstack@citrix.com.\nFor debugging suggestions, see https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures\n\n- check-citrix-xenserver http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/7/24584/index.html : SUCCESS in 01 h 09 m 26 s","accounts_in_message":[],"_revision_number":7},{"id":"f4e574dfe652d7164454ed9c1e629ca07b7597e1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 22:42:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/7/check/gate-nova-docs/27eb626//doc/build/html/ : SUCCESS in 4m 52s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/7/check/gate-nova-pep8/ac5c2d3/ : SUCCESS in 4m 45s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/7/check/gate-nova-python27-db/6acdf88/ : SUCCESS in 12m 05s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/7/check/gate-nova-python34-db/9f984e4/ : SUCCESS in 11m 05s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/7/check/gate-nova-python35-db-nv/629d415/ : SUCCESS in 12m 54s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-full/4a9b478/ : SUCCESS in 35m 44s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-postgres-full/eacec5c/ : SUCCESS in 40m 40s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-neutron-full/cca5d3f/ : SUCCESS in 1h 04m 23s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/7/check/gate-grenade-dsvm/0099709/ : SUCCESS in 28m 24s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/7/check/gate-nova-releasenotes/bd5792f//releasenotes/build/html/ : SUCCESS in 3m 37s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/7/check/gate-nova-tox-db-functional/a66d9ef/ : SUCCESS in 13m 46s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/7/check/gate-grenade-dsvm-multinode/fa278d2/ : SUCCESS in 44m 54s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/cd01e42/ : SUCCESS in 36m 02s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-cells/9423a4a/ : SUCCESS in 39m 34s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph/455208f/ : SUCCESS in 58m 09s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-multinode-full/f8a9e0f/ : SUCCESS in 58m 55s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/d0adeef/ : SUCCESS in 1h 09m 33s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/7/check/gate-tempest-dsvm-multinode-live-migration/4f37b0b/ : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"55a8882a76d6f8685e1031f9fc722d87fe5b7188","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-13 22:44:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/7/check/dsvm-tempest-xen/e2d4616 : FAILURE in 1h 20m 11s","accounts_in_message":[],"_revision_number":7},{"id":"e0c0f6aee90d3a73ae257fb4b54d9a6531944d34","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 22:56:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline). To recheck leave a comment with intel-nfv-ci recheck. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel-NFV-CI.\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/7/check/tempest-dsvm-full-nfv/9d59ba5 : FAILURE in 1h 41m 09s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/7/check/tempest-dsvm-intel-nfv/419e29b : SUCCESS in 36m 23s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/7/check/tempest-dsvm-ovsdpdk-nfv-networking/b54ca9c : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":7},{"id":"840493628b9283047c1c0a5980129d685b219083","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-13 23:51:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/7/ : FAILURE in 2h 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e166d3eb27f4ed8b213d41953c618cef66835eae","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-07-14 04:12:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/7/check/dsvm-tempest-kvm/c06e0c2 : FAILURE in 1h 15m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"8ed2d6c98398b521cb1b5800849d4be2e9d9ce8c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-07-14 12:35:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/7/check/check-dsvm-tempest-vz7-exe-minimal/5963fc0 : SUCCESS in 1h 25m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"4d1e8f60fed806767853318ea55dc9ee1fdd00d6","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-14 14:34:36.000000000","message":"Patch Set 7:\n\nxen: recheck","accounts_in_message":[],"_revision_number":7},{"id":"566a8228bc03d05890cf18a22de7f6efe35f643f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-07-14 16:00:30.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nCode looks good.\n\nI wonder about doing the direct import, but that means changing the service API. We can do that in a follow up if needed.\n\nSlight niggle on the format of the reno note.","accounts_in_message":[],"_revision_number":7},{"id":"fcc5941836b0a7d3f1e7f80c6d0882b4f3c67bdf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-14 16:42:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/7/check/dsvm-tempest-xen/d0cadf8 : SUCCESS in 1h 32m 12s","accounts_in_message":[],"_revision_number":7},{"id":"a7226896a2d241f5ce0be8173585b5419154d812","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-14 17:44:16.000000000","message":"Patch Set 8: Published edit on patch set 7","accounts_in_message":[],"_revision_number":8},{"id":"04392e4b02a1266cdc2a0c8c64a8d3659be5849a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-14 18:51:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs http://docs-draft.openstack.org/53/338553/8/check/gate-nova-docs/84e2c86//doc/build/html/ : SUCCESS in 3m 18s\n- gate-nova-pep8 http://logs.openstack.org/53/338553/8/check/gate-nova-pep8/8a14019/ : SUCCESS in 6m 08s\n- gate-nova-python27-db http://logs.openstack.org/53/338553/8/check/gate-nova-python27-db/9be5aa1/ : SUCCESS in 11m 00s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/8/check/gate-nova-python34-db/478f037/ : SUCCESS in 9m 00s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/8/check/gate-nova-python35-db-nv/abe0dbc/ : SUCCESS in 14m 38s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-full/4acb3f1/ : SUCCESS in 41m 48s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-postgres-full/00e4d4e/ : SUCCESS in 39m 22s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-neutron-full/afaa565/ : SUCCESS in 1h 01m 02s\n- gate-grenade-dsvm http://logs.openstack.org/53/338553/8/check/gate-grenade-dsvm/ca57755/ : SUCCESS in 35m 28s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/8/check/gate-nova-releasenotes/366a0f0//releasenotes/build/html/ : SUCCESS in 5m 53s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/8/check/gate-nova-tox-db-functional/c8abbcd/ : SUCCESS in 14m 34s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/8/check/gate-grenade-dsvm-multinode/cee07a7/ : SUCCESS in 28m 58s\n- gate-tempest-dsvm-ironic-pxe_ipa-nv http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-ironic-pxe_ipa-nv/59fca01/ : SUCCESS in 27m 51s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-cells/7878ad9/ : SUCCESS in 39m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph/873e165/ : SUCCESS in 50m 21s\n- gate-tempest-dsvm-multinode-full http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-multinode-full/1e8defa/ : FAILURE in 1h 00m 46s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/8/check/gate-tempest-dsvm-multinode-live-migration/b512633/ : SUCCESS in 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b55cafbeafc9aa133b1499491914e330ef948d32","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-07-14 19:33:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"031e451683d457ad5fc8d82c1625fdc778d95993","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-14 20:05:09.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/338553/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"3b3971e179c7a4fc85aa771e7800a2fc05e4f61a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-07-14 20:42:48.000000000","message":"Patch Set 8:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/8/check/check-ibm-tempest-dsvm-full/2cdabce/ : SUCCESS in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7755453fb258bd488bff32c41517eda5af9bd62f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-07-15 00:11:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/8/check/dsvm-tempest-xen/cb87476 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":8},{"id":"1695313d8e55688623e87dad0ecbea261694b0ba","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-15 00:32:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/8/check/tempest-dsvm-full-nfv/24727a7 : SUCCESS in 1h 32m 21s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/8/check/tempest-dsvm-intel-nfv/fb4a6b0 : SUCCESS in 34m 46s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/8/check/tempest-dsvm-ovsdpdk-nfv-networking/1568a78 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":8},{"id":"a894c5e32ba93458510abc81389f9703de3b13f5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-07-15 01:15:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/8/check/dsvm-tempest-nova-network/8dbf26e : SUCCESS in 1h 09m 41s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/8/check/dsvm-tempest-neutron-network/036e3bd : SUCCESS in 1h 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1006fb620cd2df9d8efccd27afaacfaf496cbdd8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-07-15 01:20:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/8/check/dsvm-tempest-kvm/7aaa936 : FAILURE in 1h 28m 59s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"23384f905e26cfe501ec6209b32e05923d54729a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-07-15 04:37:52.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"0543e4d20072cc504fe5559b1ce1ad544744ff85","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-07-15 07:36:32.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"88f70c8e1ba3822fe29bd22b300cd8a33ca288db","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-07-15 10:48:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/8/check/check-dsvm-tempest-vz7-exe-minimal/a56309a : SUCCESS in 2h 09m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"f92bb9598409a5d7a499923c8f75c6ccbd0969e3","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-07-20 14:37:41.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"54efccb2e0527c8f635da38ea7886ff197ccb295","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2016-07-20 19:22:38.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d6481b01f58172a4f0991d2457d3fcc340f653c3","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-08-09 10:39:36.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dc0967a361d2bac0200db1e0a513a5821d39d041","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-08-09 10:40:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"521fc6245417622c964e996ca7ee3727566d19a1","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-08-09 10:40:13.000000000","message":"Topic changed from deprecate-config-options-service to bp/centralize-config-options-newton","accounts_in_message":[],"_revision_number":8},{"id":"985d6a88af0e4e7013c8c6408c8655bf64da6777","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-08-15 16:56:00.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"42e9bff6e94c05cca7cf7fda79c57c08216d32a6","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-08-15 17:01:51.000000000","message":"Patch Set 8: Code-Review-1\n\nThis patch needs to rebase from latest of https://review.openstack.org/#/c/332861 and also add bp in commit msg for linking.","accounts_in_message":[],"_revision_number":8},{"id":"bb8e658ca25270a55e59f0a1f4227d4794396ce1","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-08-15 18:39:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ea56d55f2cbbfbeb2b27dc37e3eed1a5ad216fd2","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-08-15 18:40:22.000000000","message":"Patch Set 10: Published edit on patch set 9","accounts_in_message":[],"_revision_number":10},{"id":"0db0ad44f6293bf3023bf58f5cbab46e578e3737","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-08-15 18:41:06.000000000","message":"Patch Set 11: Published edit on patch set 10","accounts_in_message":[],"_revision_number":11},{"id":"1c08130883a8559eca0effe89daf902afbeb907e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-15 19:06:41.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/338553/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"7085dfc2f66db5616b63827e952df186cea1b130","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-15 19:17:53.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/338553/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"36c0f58e87e6f966654bafab188ec0f86bddaf27","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-15 19:19:31.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/pci/338553/11 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"a46e7b50de0886fca6ca373173c6fcfa6534df7d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-08-15 19:40:12.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/11/check-nova/Nova-ML2-Sriov/7ffb570 : SUCCESS in 32m 36s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/11/check-nova/Nova-MACVTAP-ML2-Sriov/af55fed : SUCCESS in 31m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"685d88f76831d666b38a064e4a1295cc518d1819","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-08-15 19:48:33.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/11/check/check-ibm-tempest-dsvm-full/a620d3a/ : SUCCESS in 1h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a2e55f16bfa382d4a23c3d23896990d1a7bd918a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-08-15 20:06:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/11/ : FAILURE in 1h 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7aab77a16974f43b88a22c231dd422b9240991f8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-15 21:02:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/53/338553/11/check/tempest-dsvm-full-nfv/9468320 : SUCCESS in 1h 46m 48s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/53/338553/11/check/tempest-dsvm-intel-nfv/ff395ef : SUCCESS in 35m 38s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/53/338553/11/check/tempest-dsvm-ovsdpdk-nfv-networking/c2b60ef : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":11},{"id":"8d9b911afda85b94ad37820b24ba6a475ebc4a8e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-08-15 21:24:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/11/check/dsvm-tempest-nova-network/6874675 : SUCCESS in 1h 55m 09s","accounts_in_message":[],"_revision_number":11},{"id":"bd789136157dbf5fa1a176dbd4b0a3c0449cb1ab","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-08-15 22:38:47.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/11/check/dsvm-tempest-kvm/8804eb4 : FAILURE in 1h 25m 41s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"ca4a5324774ae41a0079fb02e100c9df6f528011","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-15 23:25:51.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/11/check/gate-nova-docs-ubuntu-xenial/5666a00//doc/build/html/ : SUCCESS in 4m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/11/check/gate-nova-pep8-ubuntu-xenial/fba2c46/ : FAILURE in 39s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/11/check/gate-nova-python27-db-ubuntu-xenial/23c7770/ : FAILURE in 20m 00s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/11/check/gate-nova-python34-db/1e58caa/ : FAILURE in 12m 08s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/11/check/gate-nova-python35-db-nv/88d8257/ : FAILURE in 16m 08s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/53/338553/11/check/gate-grenade-dsvm-neutron-ubuntu-trusty/3781750/ : SUCCESS in 1h 12m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/615c06a/ : SUCCESS in 1h 18m 48s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-full-ubuntu-xenial/74af6ce/ : SUCCESS in 1h 00m 48s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/8a59597/ : SUCCESS in 1h 06m 59s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/53/338553/11/check/gate-grenade-dsvm-ubuntu-trusty/5d3e0ca/ : SUCCESS in 1h 10m 18s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/11/check/gate-nova-releasenotes/e7aa635//releasenotes/build/html/ : SUCCESS in 4m 50s\n- gate-nova-tox-db-functional http://logs.openstack.org/53/338553/11/check/gate-nova-tox-db-functional/5835bc6/ : SUCCESS in 16m 40s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/11/check/gate-grenade-dsvm-multinode/f5b0fe1/ : SUCCESS in 58m 40s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/2270d47/ : FAILURE in 34m 25s (non-voting)\n- gate-tempest-dsvm-cells http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-cells/c1e6add/ : FAILURE in 24m 24s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-full-devstack-plugin-ceph/1fc4ab2/ : SUCCESS in 1h 15m 35s\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-neutron-multinode-full/ec9125c/ : SUCCESS in 2h 42m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/53/338553/11/check/gate-tempest-dsvm-multinode-live-migration/9e09053/ : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"adbeadd94985429dcaaee846974090640b4a46ee","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-08-15 23:40:55.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/11/check/check-dsvm-tempest-vz7-exe-minimal/2b3d7c2 : SUCCESS in 1h 15m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"fd497beabcfa18ee4b438b79da4b816e0049ffd6","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-08-16 00:36:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/11/check/dsvm-tempest-xen/fd32202 : SUCCESS in 1h 45m 42s","accounts_in_message":[],"_revision_number":11},{"id":"e449a759f6400b825258c372bc61fbb49d7b3ace","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-08-16 09:23:32.000000000","message":"Patch Set 11: Code-Review-2\n\nLets save this for next cycle now. This is a bit intrusive.","accounts_in_message":[],"_revision_number":11},{"id":"916c252d0c2d7530b581c414451445480995e430","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-08-16 09:25:40.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/11/68/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"2dae6fc6e4abea5f1c14c19f487830a2877c2aed","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-09-14 08:15:51.000000000","message":"Patch Set 12: Published edit on patch set 11","accounts_in_message":[],"_revision_number":12},{"id":"5df9c3c4e4bb844ab674cce8ac98b1dfccbc170b","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-09-14 08:16:01.000000000","message":"Topic changed from bp/centralize-config-options-newton to bp/centralize-config-options-ocata","accounts_in_message":[],"_revision_number":12},{"id":"aaa639adcdbda668586ccafc20ba1281a2a99ecb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-14 09:16:45.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/12/check/check-ibm-tempest-dsvm-full/c7c7bc9/ : SUCCESS in 1h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b2596e123e04c0e1350b5d2753268887693f3c64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-14 09:42:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/12/check/gate-nova-docs-ubuntu-xenial/1273384//doc/build/html/ : SUCCESS in 3m 39s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-nova-pep8-ubuntu-xenial/8a468fd/ : SUCCESS in 5m 03s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-nova-python27-db-ubuntu-xenial/efab9a4/ : FAILURE in 16m 39s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/12/check/gate-nova-python34-db/1e8e473/ : FAILURE in 12m 21s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/12/check/gate-nova-python35-db-nv/8bd6a68/ : FAILURE in 13m 14s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/53/338553/12/check/gate-grenade-dsvm-neutron-ubuntu-trusty/b0e3899/ : SUCCESS in 47m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f1a0ffc/ : SUCCESS in 1h 04m 11s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-full-ubuntu-xenial/46524bc/ : SUCCESS in 53m 19s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/53/338553/12/check/gate-grenade-dsvm-ubuntu-trusty/fd4ad5a/ : SUCCESS in 24m 44s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/12/check/gate-nova-releasenotes/2b15782//releasenotes/build/html/ : SUCCESS in 2m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-nova-tox-db-functional-ubuntu-xenial/fd48107/ : SUCCESS in 18m 51s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/12/check/gate-grenade-dsvm-multinode/eb4534b/ : SUCCESS in 44m 14s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-cells-ubuntu-xenial/fef70e6/ : SUCCESS in 37m 50s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/224e7ed/ : SUCCESS in 1h 00m 07s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/be76b82/ : SUCCESS in 1h 25m 30s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/0ecfb00/ : SUCCESS in 1h 04m 35s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/bb59f32/ : SUCCESS in 17m 20s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/12/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/52d70a3/ : SUCCESS in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9843a835b363b7d1bd21f13fe6fd01362bf8488","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-14 11:26:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/12/check/dsvm-tempest-xen/e0162ce : SUCCESS in 1h 39m 10s","accounts_in_message":[],"_revision_number":12},{"id":"94f5adbc18f9b52b830ce4b0957e93489c683a8b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-14 11:48:36.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/12/ : FAILURE in 2h 35m 56s","accounts_in_message":[],"_revision_number":12},{"id":"48ca4f4880a3a9508459d325a2247bff9a731258","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-14 12:30:41.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/12/check-nova/Nova-ML2-Sriov/8e542cb : SUCCESS in 32m 34s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/12/check-nova/Nova-MACVTAP-ML2-Sriov/62a4184 : SUCCESS in 36m 54s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":12},{"id":"0c12628b7ee7e0c16b6087a6fec40ea0f68c8a33","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-14 12:31:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/12/check/dsvm-tempest-nova-network/815b4a7 : SUCCESS in 1h 20m 27s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/12/check/dsvm-tempest-neutron-network/6e96ffc : SUCCESS in 1h 58m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c529839134c0e7e746e7a92c8064b366d753d294","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-14 13:21:07.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/12/check/check-dsvm-tempest-vz7-exe-minimal/6fedf53 : SUCCESS in 1h 24m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"eed6d8234b834dc1f9532a09ba5d75560bb71195","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-14 13:54:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/12/check/tempest-dsvm-full-nfv-xenial/7ae8f78 : FAILURE in 2h 12m 12s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/12/check/tempest-dsvm-intel-nfv-xenial/b72b6c0 : SUCCESS in 49m 30s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5a814f9 : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":12},{"id":"616e21a26fae5f1e70c3aa593f9c6c35ba1188a2","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-14 16:16:43.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/12/check/dsvm-tempest-kvm/0c9f327 : SUCCESS in 1h 25m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"5820543338143b25ee6a3f466183ee41eca26eca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-14 18:57:29.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/12/1031/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"e3858762b08c914c1b938af1ba158dc6338725bf","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-09-20 15:26:40.000000000","message":"Patch Set 12: -Code-Review\n\ncool, its next cycle now, so we can look at this one again.","accounts_in_message":[],"_revision_number":12},{"id":"bb72cc052f77a6a9d4c7d4a1e529513d2cbec0b7","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-09-23 22:29:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"bfb41050930602ceedba911ebc50c34f3e9da3f0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-23 22:46:57.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/338553/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"4f70f18db5b12768cedab610b2211e47f3a01f3d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-23 23:33:22.000000000","message":"Patch Set 13:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/13/check/check-ibm-tempest-dsvm-full/b7d9a7c/ : SUCCESS in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"43ccbe91893dcee81abf2290fe3d2cdefc1bb492","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-23 23:50:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/13/check/gate-nova-docs-ubuntu-xenial/f31b670//doc/build/html/ : SUCCESS in 2m 28s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-nova-pep8-ubuntu-xenial/3e79ff7/ : SUCCESS in 5m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-nova-python27-db-ubuntu-xenial/864c6bd/ : SUCCESS in 14m 16s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/13/check/gate-nova-python34-db/46aaef5/ : SUCCESS in 12m 08s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/13/check/gate-nova-python35-db-nv/a1be0a8/ : SUCCESS in 12m 25s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/57c7bf9/ : SUCCESS in 1h 02m 55s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-grenade-dsvm-neutron-ubuntu-xenial/68d4f03/ : SUCCESS in 55m 49s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-full-ubuntu-xenial/d13dc9f/ : SUCCESS in 53m 32s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-grenade-dsvm-ubuntu-xenial/8b649b7/ : SUCCESS in 36m 54s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/13/check/gate-nova-releasenotes/7847241//releasenotes/build/html/ : SUCCESS in 3m 33s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-nova-tox-db-functional-ubuntu-xenial/9a6a0aa/ : SUCCESS in 12m 46s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/13/check/gate-grenade-dsvm-multinode/13ebfcc/ : SUCCESS in 49m 54s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-cells-ubuntu-xenial/25a497e/ : SUCCESS in 36m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/162fd6e/ : SUCCESS in 53m 28s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/c600753/ : SUCCESS in 1h 14m 40s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/fe4f2a0/ : SUCCESS in 1h 05m 21s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b773f68/ : SUCCESS in 29m 56s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/13/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/3f20e5d/ : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0223e7727f4271671a7c2282b7f22a239f4bbde8","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-23 23:55:18.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/13/check-nova/Nova-ML2-Sriov/c978ae7 : SUCCESS in 33m 35s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/13/check-nova/Nova-MACVTAP-ML2-Sriov/c0fbd8e : SUCCESS in 33m 44s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"0d26404c095a52b3463cfd7ca38c76a419ea695e","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-24 00:22:28.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/13/ : SUCCESS in 1h 52m 21s","accounts_in_message":[],"_revision_number":13},{"id":"02f3149bbc5ce7d6dee63745b710e1f96808b2f8","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-24 00:25:16.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/13/check/dsvm-tempest-xen/de6e451 : FAILURE in 1h 47m 55s","accounts_in_message":[],"_revision_number":13},{"id":"5ec7e4aaaacce29b20130a4e6ee2e35896486db7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-24 00:59:42.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"8556dc96b110d766a1675a6d1db48a49f020e49c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-24 01:29:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/13/check/dsvm-tempest-nova-network/cdeaacc : SUCCESS in 1h 30m 04s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/13/check/dsvm-tempest-neutron-network/2df9bdb : SUCCESS in 1h 43m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"402220f486aac78d5f7a6c4e557639a0e32f8233","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-24 01:40:55.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/13/check/tempest-dsvm-full-nfv-xenial/8191bf5 : SUCCESS in 1h 48m 29s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/13/check/tempest-dsvm-intel-nfv-xenial/e3e49c4 : SUCCESS in 36m 36s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/13/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/3b1b76a : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":13},{"id":"167e68f64c5a518efde405019a11909cb765eef8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-27 12:11:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/13/1342/ : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"41d738c4f2457a8b17ca2026b3dca528e9e59074","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-09-27 12:25:42.000000000","message":"Patch Set 13:\n\nxen: recheck","accounts_in_message":[],"_revision_number":13},{"id":"1f36a04bcd0f591c662603b8125f6fa437366251","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-09-27 13:56:26.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2b21c1fb21a05aa3d58c0b2ee95558dcd1df4f4e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-27 14:24:35.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/13/check/dsvm-tempest-xen/aa01314 : FAILURE in 1h 47m 34s","accounts_in_message":[],"_revision_number":13},{"id":"f52907602d9d73c1cadfc21d4849bc3b92f616f4","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-09-27 16:20:29.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c2892eee07cd0e39030c68adb0c5d5307e45160e","author":{"_account_id":23312,"name":"kangyufei","email":"kangyf@inspur.com","username":"kangyufei"},"date":"2016-09-28 08:59:05.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c1d44cbf383e1a85469ac7fa724308165088c3b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-09-28 15:44:58.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)\n\nI don\u0027t like holding this up for what are nits, but I do think we should be consistent in release notes to ease cognitive load on folks","accounts_in_message":[],"_revision_number":13},{"id":"e00e73761bcfaa2e3fe1b0d43ab51e576c9b1968","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-09-28 21:14:08.000000000","message":"Patch Set 14: Published edit on patch set 13","accounts_in_message":[],"_revision_number":14},{"id":"9db7ad2ce87367cacf0a9e431ddb168bad7d757c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-28 21:36:13.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/338553/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"2be7d3f3c98b8f7c3af3517ea64d49350cdc55a2","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-28 22:52:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/14/check/dsvm-tempest-xen/3b5edb9 : SUCCESS in 1h 37m 58s","accounts_in_message":[],"_revision_number":14},{"id":"683b461630551efd50cd2b87b58c32b41668267d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-28 23:28:42.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/14/check/gate-nova-docs-ubuntu-xenial/b644f05//doc/build/html/ : SUCCESS in 6m 27s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-nova-pep8-ubuntu-xenial/753e8ee/ : SUCCESS in 8m 54s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-nova-python27-db-ubuntu-xenial/64fd2b8/ : SUCCESS in 17m 17s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/14/check/gate-nova-python34-db/72337b8/ : SUCCESS in 15m 27s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/14/check/gate-nova-python35-db-nv/41469a8/ : SUCCESS in 17m 11s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7279ccc/ : SUCCESS in 46m 05s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-grenade-dsvm-neutron-ubuntu-xenial/9d0adec/ : SUCCESS in 32m 10s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-full-ubuntu-xenial/5300890/ : SUCCESS in 40m 28s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-grenade-dsvm-ubuntu-xenial/06a4408/ : SUCCESS in 26m 29s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/14/check/gate-nova-releasenotes/6f8ccfa//releasenotes/build/html/ : SUCCESS in 7m 47s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-nova-tox-db-functional-ubuntu-xenial/6ce2557/ : FAILURE in 21m 03s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/14/check/gate-grenade-dsvm-multinode/f43415a/ : SUCCESS in 30m 15s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-cells-ubuntu-xenial/420318d/ : SUCCESS in 47m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/c9dfb99/ : SUCCESS in 59m 45s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/7ace760/ : FAILURE in 25m 12s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/b1e6994/ : SUCCESS in 1h 07m 09s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/653bc2e/ : SUCCESS in 25m 14s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/14/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/ab1c312/ : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"fb4f2d94a45a4c6b291c22395fc8e7a167b8f9eb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-29 00:56:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/14/check/dsvm-tempest-nova-network/f79a29d : SUCCESS in 1h 31m 01s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/14/check/dsvm-tempest-neutron-network/92a45d2 : SUCCESS in 1h 57m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"979d21cd97c53f9a9c35aa9c3304c8e240166f39","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-29 03:11:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/14/check/tempest-dsvm-full-nfv-xenial/04e7acd : SUCCESS in 2h 04m 36s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/14/check/tempest-dsvm-intel-nfv-xenial/2482d0d : SUCCESS in 35m 35s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/14/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/842a7c3 : SUCCESS in 52m 35s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"28f29f88e369afdbc193b0a76a0fa771d26d78e1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-29 15:34:55.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/14/ : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":14},{"id":"308d84e32465e106afda283b162aa76e8e403555","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-09-29 17:44:19.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"fbdd4e5458dc0ca5dee6d7b868438020a44c1849","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-29 17:44:53.000000000","message":"Patch Set 15:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c969c6235455ebf28b12bf97be0d947aa1419116","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-29 18:19:10.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/338553/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"b895fc9e11efea979710aa11fdb3bda67a3b0bdc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-29 19:08:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/15/check/gate-nova-docs-ubuntu-xenial/ca8001b//doc/build/html/ : SUCCESS in 3m 05s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-nova-pep8-ubuntu-xenial/a2a88a2/ : SUCCESS in 4m 41s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-nova-python27-db-ubuntu-xenial/7f7dbd5/ : SUCCESS in 11m 10s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/15/check/gate-nova-python34-db/48fadad/ : SUCCESS in 9m 57s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/15/check/gate-nova-python35-db-nv/bee8ac7/ : SUCCESS in 10m 24s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/cddf0a8/ : SUCCESS in 47m 28s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b23885c/ : SUCCESS in 1h 08m 27s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-full-ubuntu-xenial/2b43291/ : SUCCESS in 1h 05m 39s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-grenade-dsvm-ubuntu-xenial/2351370/ : SUCCESS in 27m 02s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/15/check/gate-nova-releasenotes/f28d2a7//releasenotes/build/html/ : SUCCESS in 5m 58s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-nova-tox-db-functional-ubuntu-xenial/ed5e522/ : SUCCESS in 12m 21s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/15/check/gate-grenade-dsvm-multinode/c614afc/ : SUCCESS in 29m 22s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-cells-ubuntu-xenial/55f9fad/ : SUCCESS in 30m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1144f36/ : SUCCESS in 42m 57s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/6ac9b3f/ : SUCCESS in 1h 20m 07s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/80b6ba4/ : SUCCESS in 46m 12s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/bb3356a/ : SUCCESS in 19m 17s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/15/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/552ef1c/ : SUCCESS in 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"96dde5b9ca9b6c19d7342ba44ab0c21329833408","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-29 19:19:26.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/15/check-nova/Nova-ML2-Sriov/6416ed3 : SUCCESS in 34m 02s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/15/check-nova/Nova-MACVTAP-ML2-Sriov/4a3880e : SUCCESS in 34m 25s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"602bac8aa2e735dd4bacd1fea469e3e956c3ceec","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-29 19:39:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/15/check/dsvm-tempest-xen/198205a : SUCCESS in 1h 41m 29s","accounts_in_message":[],"_revision_number":15},{"id":"a20a34f40c88b69372fb03938e6cc1d1a99fe4f1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-29 19:53:44.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/15/check/dsvm-tempest-nova-network/6ecc981 : SUCCESS in 1h 53m 23s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/15/check/dsvm-tempest-neutron-network/7cb398f : SUCCESS in 1h 57m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d406dcd3c24697f900def63f2a682aa08645b65d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-29 20:23:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/15/check/tempest-dsvm-full-nfv-xenial/3388583 : SUCCESS in 1h 46m 11s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/15/check/tempest-dsvm-intel-nfv-xenial/b448c50 : SUCCESS in 36m 08s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c100e3a : SUCCESS in 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"91693fe8a3d2e94b4ccab4838c340dfc356168f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-29 20:47:46.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"030e7f81ceadd163f10fc3931c91c27f09efdd4d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-30 01:26:46.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/15/ : FAILURE in 2h 52m 51s","accounts_in_message":[],"_revision_number":15},{"id":"34e26ddaf84456ceb18a7c4f07e0918383ea4501","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-30 03:06:28.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/13/1501/ : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"3c93b869b73cad28427b3da5b97274884fc47b13","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-30 12:53:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/15/1549/ : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"2770ab6a8a0137766a70a3a4d52c1794af278b75","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-30 20:23:37.000000000","message":"Patch Set 14: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/338553/14/1587/ : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"f5f39eb1d254a8fc154a9baf7e73024e088e153a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-10-03 13:50:24.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\nGreat work, this is really close.\n\nI just have one worry, from a long term maintenance point of view. I don\u0027t like all those tests having to know the name of the manager. Its almost tempting to do that in a follow up patch, but we would probably forget about it if we don\u0027t do it now.\n\nGiven there is an niggle here, its tempting to include the conductor manager in the same patch, given it looks like most of the lines that will change are sat next to each other.\n\nI am tempted to say lets change all these to use a new method first, then do this deprecation patch on top of that. Otherwise this change might turn into a bit of a monster.","accounts_in_message":[],"_revision_number":15},{"id":"8f58bd78b0cad9613169bbcd19b63a0aa4dacc82","author":{"_account_id":10618,"name":"Pavel Kholkin","email":"p.v.holkin@mail.ru","username":"pkholkin"},"date":"2016-10-06 14:10:46.000000000","message":"Patch Set 15: Code-Review-1\n\n(3 comments)\n\njust small style issues","accounts_in_message":[],"_revision_number":15},{"id":"aa26b2e1326b80e9c018b9b3000be7357c9fe0d8","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-06 21:30:53.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"efe45b4037d4ab97ba98332c4df045518fecad7d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-10-07 16:13:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"da16ce847b8ecda8a8323d35b11508d85e51d052","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-07 21:09:54.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2cd6d3857b74df1edeaaadf4f8895a7f793df2e9","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-07 21:12:10.000000000","message":"Patch Set 17: Published edit on patch set 16","accounts_in_message":[],"_revision_number":17},{"id":"66b752db6181cdac901bac4dbbe7128d95f59164","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-07 21:28:04.000000000","message":"Patch Set 17:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/17/check/check-ibm-tempest-dsvm-full/3daaae5/ : FAILURE in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"dd1806f044f981271b9d17bd7379d15f497bc150","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-10-07 21:31:56.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/pci/338553/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"f3304bd77e4d7a113e086c2c9d7136887f8054f1","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-10-07 21:34:18.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/338553/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"69f94cc3b8ce094507cd7e8bbe3b977c4878796f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-10-07 22:04:03.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/17/check/check-dsvm-tempest-vz7-exe-minimal/08cf593 : SUCCESS in 51m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"3a478fccd64657a131bcce2327ddc11ee4de215c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-10-07 22:57:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/17/check/dsvm-tempest-xen/f3aa2ac : FAILURE in 1h 35m 44s","accounts_in_message":[],"_revision_number":17},{"id":"69c0bd1b882a5d14bd2286a6116279b43b1a8048","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-07 23:02:18.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/17/check/gate-nova-docs-ubuntu-xenial/01b9f2c//doc/build/html/ : SUCCESS in 2m 49s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-nova-pep8-ubuntu-xenial/0dbcd55/ : FAILURE in 4m 41s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-nova-python27-db-ubuntu-xenial/7d3ee08/ : FAILURE in 14m 49s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/17/check/gate-nova-python34-db/380a49a/ : FAILURE in 7m 52s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/17/check/gate-nova-python35-db-nv/237d116/ : FAILURE in 13m 56s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7668315/ : SUCCESS in 1h 03m 40s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a203037/ : FAILURE in 25m 57s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-full-ubuntu-xenial/a4e6d72/ : SUCCESS in 36m 47s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-grenade-dsvm-ubuntu-xenial/a3ef419/ : SUCCESS in 37m 19s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/17/check/gate-nova-releasenotes/0e1d721//releasenotes/build/html/ : SUCCESS in 3m 52s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-nova-tox-db-functional-ubuntu-xenial/01b5fcb/ : SUCCESS in 11m 26s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/17/check/gate-grenade-dsvm-multinode/1eb8447/ : SUCCESS in 48m 47s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/17/check/gate-grenade-dsvm-multinode-live-migration-nv/f9bb22c/ : SUCCESS in 30m 43s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-cells-ubuntu-xenial/0a0d8ca/ : SUCCESS in 30m 52s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/c90911d/ : SUCCESS in 50m 32s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/2be5b23/ : SUCCESS in 1h 22m 13s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/806052c/ : SUCCESS in 1h 07m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/950ab1f/ : SUCCESS in 31m 28s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/17/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/31b2da7/ : SUCCESS in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"95fa63640e42e1746bb25553897c2149b33af0d9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-07 23:05:47.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/17/ : SUCCESS in 1h 53m 09s","accounts_in_message":[],"_revision_number":17},{"id":"b4f46e2c91eb5f81e6ec6d6c0a9822becff8bc02","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-07 23:15:44.000000000","message":"Patch Set 17:\n\ncheck hyper-v","accounts_in_message":[],"_revision_number":17},{"id":"3cf7b3f5daec3b66753837eab029524707828c7d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-10-07 23:42:39.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/17/check-nova/Nova-ML2-Sriov/b5c670a : SUCCESS in 29m 26s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/17/check-nova/Nova-MACVTAP-ML2-Sriov/f4c54d9 : SUCCESS in 34m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"20e1f31a40b581b68147eb5edd43d5c300d6e0f1","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-08 01:15:30.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/17/ : SUCCESS in 1h 59m 25s","accounts_in_message":[],"_revision_number":17},{"id":"f263a2db31923f0fcdfb87727b429bd525497d70","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-10-08 02:28:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/17/check/dsvm-tempest-kvm/5fbc3a0 : SUCCESS in 1h 21m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"42ef7835da9d5843a984e082119c9c713d82f404","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-08 07:15:52.000000000","message":"Patch Set 17: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/338553/17/256/ : SUCCESS\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":17},{"id":"a52791daba867089415589f610101812220bdf4d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-10-08 09:53:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/17/check/dsvm-tempest-nova-network/944551a : SUCCESS in 1h 45m 35s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/17/check/dsvm-tempest-neutron-network/8ec9950 : SUCCESS in 1h 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3b59d0ba009af9f330bf3a3aaf05149ef014fe08","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-10-08 18:38:03.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/17/check/tempest-dsvm-full-nfv-xenial/985c848 : SUCCESS in 1h 38m 12s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/17/check/tempest-dsvm-intel-nfv-xenial/22712d2 : SUCCESS in 35m 20s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/17/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/00c0935 : SUCCESS in 53m 47s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/17/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/db9e5ef : FAILURE in 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"410b2fd65d143b127f0e4faecd4c6b3d5652d2ac","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-10-10 10:02:45.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1247ce057fa5b2b91c56ca2bee01cacf45dc33a8","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-10 15:47:33.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"02b6ce9c565e3f078cbdf52990b25de2f467f2d3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-10-10 16:16:08.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/338553/18 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"c0a9b3b25b464d07b284f1dec13c9fa661a191c8","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-10 16:22:48.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/18/ : FAILURE in 34m 49s","accounts_in_message":[],"_revision_number":18},{"id":"55fba005db778e28da0c2559b9b9946992058214","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-10 16:46:35.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/18/check/gate-nova-docs-ubuntu-xenial/10fddc0//doc/build/html/ : SUCCESS in 4m 05s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-pep8-ubuntu-xenial/9ff6d04/ : SUCCESS in 7m 13s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-python27-db-ubuntu-xenial/8a8b969/ : SUCCESS in 9m 35s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/18/check/gate-nova-python34-db/8175291/ : SUCCESS in 11m 40s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/18/check/gate-nova-python35-db-nv/94c76b3/ : SUCCESS in 8m 33s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ebb66b4/ : SUCCESS in 45m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b279738/ : SUCCESS in 32m 03s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-ubuntu-xenial/ddd0cd5/ : SUCCESS in 58m 05s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-ubuntu-xenial/6c2488c/ : SUCCESS in 25m 04s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/18/check/gate-nova-releasenotes/c870564//releasenotes/build/html/ : SUCCESS in 3m 33s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-tox-db-functional-ubuntu-xenial/7d667e5/ : SUCCESS in 15m 10s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode/b141b28/ : SUCCESS in 28m 03s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode-live-migration-nv/c410908/ : SUCCESS in 30m 44s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/6f2b4e1/ : SUCCESS in 45m 38s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/f2fc363/ : FAILURE in 54m 40s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/ed15866/ : FAILURE in 20m 33s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/1053203/ : SUCCESS in 43m 18s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/f7b1c01/ : SUCCESS in 17m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/c91cfae/ : SUCCESS in 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"29388fc8f827eca7067ed8a39d7ab3f48927b8d3","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-10-10 16:46:47.000000000","message":"Patch Set 18:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1449136\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":18},{"id":"936ac9f351ddc4e82e0eee5fafee4a631fee9430","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-10 16:48:48.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/18/check/check-ibm-tempest-dsvm-full/9df3f2d/ : SUCCESS in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4e166e3145284ab113f8000a0dd2852f79dd14d7","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-10 17:12:29.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"ef331cab65c7de6e69c9c445e6d421eccf03d5ea","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-10-10 17:37:18.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/18/check/dsvm-tempest-kvm/c675faf : SUCCESS in 1h 30m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"f670209ff1095b14f524c9f3e918cdb0532d41d3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-10-10 17:37:59.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/18/check/dsvm-tempest-xen/8babfe1 : FAILURE in 1h 47m 40s","accounts_in_message":[],"_revision_number":18},{"id":"c027813f04a27affac8ea8a93c385c854ca489e7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-10-10 17:38:23.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/18/check/dsvm-tempest-nova-network/718b938 : SUCCESS in 1h 20m 21s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/18/check/dsvm-tempest-neutron-network/d7a5f6e : SUCCESS in 1h 50m 15s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"440ae3bb795619330a5c8789110ad9f8f2e69386","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-10 17:44:40.000000000","message":"Patch Set 18:\n\nxen: recheck","accounts_in_message":[],"_revision_number":18},{"id":"b62668d24688e855d6773340ec922f604866c765","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-10 18:11:51.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-docs-ubuntu-xenial/d4d03a8/ : FAILURE in 56s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-pep8-ubuntu-xenial/1ded0d9/ : FAILURE in 1m 02s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-python27-db-ubuntu-xenial/c3d6538/ : FAILURE in 55s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/18/check/gate-nova-python34-db/4bb507c/ : SUCCESS in 11m 39s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/18/check/gate-nova-python35-db-nv/088e8df/ : FAILURE in 52s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d2bf780/ : FAILURE in 52s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8ba5649/ : SUCCESS in 31m 06s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-ubuntu-xenial/5a01a66/ : SUCCESS in 36m 34s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-ubuntu-xenial/c443e6b/ : SUCCESS in 24m 43s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/18/check/gate-nova-releasenotes/47afa50//releasenotes/build/html/ : SUCCESS in 2m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-tox-db-functional-ubuntu-xenial/3312821/ : SUCCESS in 10m 51s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode/186db0d/ : SUCCESS in 26m 17s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode-live-migration-nv/b50f180/ : FAILURE in 1m 24s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/4fd1e32/ : SUCCESS in 48m 16s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/3da5b7f/ : SUCCESS in 55m 29s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/e80d565/ : FAILURE in 19m 19s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/e99c3a7/ : SUCCESS in 57m 39s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/56f335d/ : SUCCESS in 17m 10s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/631a9e4/ : SUCCESS in 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fca37a6c23189675690df77785f6c5559bc61842","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-10 18:19:32.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/18/check/check-ibm-tempest-dsvm-full/9beb636/ : FAILURE in 1h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6f73ae270e6fb0550df8bcfa14429e0fc275b1b0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-10 18:59:39.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/18/ : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":18},{"id":"0d1488d5bfe32e92fb664e0c7b89c82d142d4868","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-10-10 19:26:39.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/18/check/dsvm-tempest-xen/612d3ea : SUCCESS in 1h 40m 32s","accounts_in_message":[],"_revision_number":18},{"id":"ab32b497dd7178a39f0ffb5d1fb4d1478a5330b1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-10 20:44:23.000000000","message":"Patch Set 18: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/338553/18/326/ : SUCCESS\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":18},{"id":"51d4da25547cb2ac27a3be4ee28a42a8a115b47f","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-10 22:17:06.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"28c8c6aeec2f2a8bbce266d903db827ff56b7dac","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-10 22:54:00.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/18/ : FAILURE in 36m 33s","accounts_in_message":[],"_revision_number":18},{"id":"1c22c6e544132766f2faa28bc3b3590cac048965","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-10 23:11:33.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-docs-ubuntu-xenial/82e9588/ : FAILURE in 4m 34s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-pep8-ubuntu-xenial/e71ff45/ : SUCCESS in 7m 08s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-python27-db-ubuntu-xenial/4dab8ff/ : SUCCESS in 13m 45s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/18/check/gate-nova-python34-db/f42bcff/ : FAILURE in 4m 27s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/18/check/gate-nova-python35-db-nv/17c2153/ : SUCCESS in 12m 06s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a65c81f/ : FAILURE in 3m 02s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4bac7ae/ : SUCCESS in 47m 36s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-ubuntu-xenial/f3dd995/ : SUCCESS in 48m 16s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-ubuntu-xenial/57af2f9/ : FAILURE in 2m 15s\n- gate-nova-releasenotes http://logs.openstack.org/53/338553/18/check/gate-nova-releasenotes/44c5b12/ : FAILURE in 3m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-tox-db-functional-ubuntu-xenial/6b2c430/ : FAILURE in 2m 12s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode/fe5702b/ : SUCCESS in 48m 15s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode-live-migration-nv/1d9a9cd/ : FAILURE in 3m 01s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/bc500e1/ : FAILURE in 1m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1ccfb42/ : FAILURE in 3m 15s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/803db20/ : FAILURE in 12m 52s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/e5c3580/ : FAILURE in 1m 44s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a0638d3/ : FAILURE in 12m 57s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/f3b2d3c/ : FAILURE in 3m 05s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a2d05db591e60dd87dc1280a25d7baa2cde1d5df","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-10-10 23:11:54.000000000","message":"Patch Set 18:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006 https://bugs.launchpad.net/bugs/1449136\n- gate-nova-docs-ubuntu-xenial: https://bugs.launchpad.net/bugs/1449136\n- gate-nova-python34-db: https://bugs.launchpad.net/bugs/1449136\n- gate-nova-releasenotes: https://bugs.launchpad.net/bugs/1449136\n- gate-nova-tox-db-functional-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-cells-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006 https://bugs.launchpad.net/bugs/1449136\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: https://bugs.launchpad.net/bugs/1449136\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006 https://bugs.launchpad.net/bugs/1449136\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":18},{"id":"c02b14dfe6910a782eaefcb1d5b5e4610ef28d8e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-10 23:16:45.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/18/check/check-ibm-tempest-dsvm-full/de92567/ : SUCCESS in 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"060bf489dd0ef60220e1085ad19c55ae21e63861","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-10 23:53:01.000000000","message":"Patch Set 18: Verified-1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/338553/18/334/ : FAILURE\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":18},{"id":"79bf40a1be971abf8e3c75f2a1e5be8a4bc08c2d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-10-11 00:14:20.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-full-nfv-xenial/fddff37 : SUCCESS in 1h 38m 46s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-intel-nfv-xenial/e7e74d4 : SUCCESS in 32m 21s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/e570285 : SUCCESS in 52m 01s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/bce291e : SUCCESS in 1h 07m 58s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e5041a6307552cefd03fa1b810a3273718df8eec","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-11 15:40:19.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"995c9a9fac071125ba61d9eacf53ddb6f8881a96","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-11 15:41:00.000000000","message":"Patch Set 18:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":18},{"id":"1718fd274de86d257e1d9298de24519b6a98a79f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-11 15:48:18.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/18/check/check-ibm-tempest-dsvm-full/9dd2af0/ : FAILURE in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e52d74407c6b9cce8e04346722671d904f01e96b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-11 16:12:53.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/18/ : FAILURE in 31m 53s","accounts_in_message":[],"_revision_number":18},{"id":"e684378cdc4379a051bbe055f38d2221c5f9416e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-11 18:36:49.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/18/check/gate-nova-docs-ubuntu-xenial/8b932c1//doc/build/html/ : SUCCESS in 3m 36s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-pep8-ubuntu-xenial/391e137/ : SUCCESS in 6m 57s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-python27-db-ubuntu-xenial/56e83f8/ : SUCCESS in 16m 35s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/18/check/gate-nova-python34-db/ab4af7b/ : SUCCESS in 13m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/18/check/gate-nova-python35-db-nv/20d1b33/ : SUCCESS in 16m 13s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e22713a/ : SUCCESS in 1h 10m 31s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a40edeb/ : SUCCESS in 59m 17s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-ubuntu-xenial/e5d1de8/ : SUCCESS in 1h 03m 52s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-ubuntu-xenial/7392c5c/ : SUCCESS in 38m 17s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/18/check/gate-nova-releasenotes/c38d929//releasenotes/build/html/ : SUCCESS in 4m 18s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-tox-db-functional-ubuntu-xenial/f156476/ : SUCCESS in 16m 20s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode/fd84aaf/ : SUCCESS in 52m 07s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode-live-migration-nv/66687a4/ : SUCCESS in 48m 05s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/8c509a3/ : SUCCESS in 43m 56s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ff80e70/ : SUCCESS in 1h 02m 37s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/0fd1ce8/ : SUCCESS in 1h 19m 59s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/886af88/ : SUCCESS in 1h 07m 31s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3fe45b0/ : FAILURE in 33m 24s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/692045b/ : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"468b8a0c2ffa033ff3e040fb6c16e4e0c26e98ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-11 21:39:02.000000000","message":"Patch Set 18:\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/338553/18/397/ : FAILURE\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":18},{"id":"1b7777efad4c2e2b4e93ba4061aa5e529e5ecfe7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-10-11 23:45:38.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-full-nfv-xenial/f085b58 : SUCCESS in 1h 36m 36s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-intel-nfv-xenial/5ca5c15 : SUCCESS in 36m 00s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/f8839b8 : SUCCESS in 52m 43s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/5b13aac : SUCCESS in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4fcc3c8056a13d1419539d07ffa601ff47ef87a4","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-12 15:06:18.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"7b8a0ee3f740746513caaa28771b8313cb4ac2eb","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-10-12 15:29:40.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/18/ : FAILURE in 21m 35s","accounts_in_message":[],"_revision_number":18},{"id":"6cb1401f93007b691829b94b62735653eee3d263","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-10-12 16:03:31.000000000","message":"Patch Set 18:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/18/check/check-ibm-tempest-dsvm-full/ec71f53/ : FAILURE in 56m 22s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6f496ae974700e4c1aa30dd8a974dfd3ff829223","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-12 16:15:26.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/18/check/gate-nova-docs-ubuntu-xenial/47ae18b//doc/build/html/ : SUCCESS in 4m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-pep8-ubuntu-xenial/746bf25/ : SUCCESS in 4m 43s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-python27-db-ubuntu-xenial/01d7498/ : SUCCESS in 13m 31s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/18/check/gate-nova-python34-db/a53f88b/ : SUCCESS in 7m 56s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/18/check/gate-nova-python35-db-nv/96346a6/ : SUCCESS in 8m 13s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/40e23f5/ : SUCCESS in 58m 38s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/741ce80/ : SUCCESS in 30m 51s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-ubuntu-xenial/7dc57c0/ : SUCCESS in 47m 29s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-ubuntu-xenial/c08574f/ : SUCCESS in 29m 27s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/18/check/gate-nova-releasenotes/6136370//releasenotes/build/html/ : SUCCESS in 4m 10s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-nova-tox-db-functional-ubuntu-xenial/7783493/ : SUCCESS in 12m 29s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode/6ac4335/ : SUCCESS in 29m 09s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/53/338553/18/check/gate-grenade-dsvm-multinode-live-migration-nv/d34b8f5/ : SUCCESS in 54m 07s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/c40a539/ : SUCCESS in 30m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5507584/ : SUCCESS in 1h 03m 21s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/5e1f191/ : FAILURE in 19m 23s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/c2df354/ : SUCCESS in 1h 07m 08s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/19e2685/ : SUCCESS in 18m 38s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/662af39/ : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"61ce2da54e48e2b2e2c5b5463fb4f774ebc1bf31","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-10-12 17:39:08.000000000","message":"Patch Set 18:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"7fc7e873e0e31470bb2ee022fcb577f50668bf52","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-12 19:13:10.000000000","message":"Patch Set 18: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/338553/18/512/ : SUCCESS\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":18},{"id":"53712648f238dff4cd3e3757d53dba76c37d6ffc","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2016-10-13 00:00:41.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"72d20dad763b23740800fd976ca314aefadd9274","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-10-13 09:55:44.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"3f93e790d714092864794e3c181e67404e1f682d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-10-13 11:04:15.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-full-nfv-xenial/bcf8a53 : SUCCESS in 1h 44m 00s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-intel-nfv-xenial/9f92cfe : SUCCESS in 36m 20s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2138091 : SUCCESS in 56m 41s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/18/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/a9a2e4c : SUCCESS in 1h 07m 14s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1a282f3c219dfca58c965a2439fe386172059c4a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-10-13 16:30:25.000000000","message":"Patch Set 18: Code-Review+2\n\n(3 comments)\n\nThanks for those updates, I think this reads better now.","accounts_in_message":[],"_revision_number":18},{"id":"138a274b71f3d8e8d3c917f88589c4960801d550","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-10-14 09:13:53.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"f52bafebe84436681bae493f39164ebf2ce2620a","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2016-10-18 11:37:58.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"20c9fa780fa0f877fd7fad371738653ce53089c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-18 13:32:48.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":18},{"id":"f0cf6f440989b49352f756c7dbf01be6934bf225","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-10-26 16:49:19.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"580a18e6f75f2feaf86ce2189b76618dcc10ecb6","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-11-09 16:24:57.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"f8210a3d48af1fe3dbf13460c41474e561f30bdc","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 17:47:42.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/19/check/dsvm-tempest-kvm/c754e04 : SUCCESS in 1h 01m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"d30197651437ab98c284e26c5da988ec19e7507a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-09 17:52:01.000000000","message":"Patch Set 19:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/19/check/tempest-dsvm-full-xenial/24a5f4e/ : SUCCESS in 1h 23m 22s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/19/check/tempest-dsvm-neutron-full-xenial/36f4d58/ : SUCCESS in 1h 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4915c391261de0339744f995ec5c4577c58d3704","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-09 18:26:24.000000000","message":"Patch Set 19:\n\n* pci-test http://52.27.155.124/pci/338553/19 : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"39a7f3de482f452ef12d867ac4c60cc3a64fcd39","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-09 18:31:59.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/19/check/dsvm-tempest-xen/1a3936a : SUCCESS in 1h 53m 35s","accounts_in_message":[],"_revision_number":19},{"id":"5ab7dd831e2b4619b5d16ae247604db1dc05009b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 18:46:49.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/19/check/gate-nova-docs-ubuntu-xenial/e510590//doc/build/html/ : SUCCESS in 2m 51s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-nova-pep8-ubuntu-xenial/1c6a449/ : SUCCESS in 4m 40s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-nova-python27-db-ubuntu-xenial/bf0797e/ : SUCCESS in 9m 03s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/19/check/gate-nova-python34-db/b9d15b8/ : SUCCESS in 15m 02s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/19/check/gate-nova-python35-db-nv/1258bbf/ : SUCCESS in 7m 56s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/851fd9c/ : SUCCESS in 46m 04s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-grenade-dsvm-neutron-ubuntu-xenial/08b4957/ : SUCCESS in 34m 47s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-grenade-dsvm-ubuntu-xenial/822eb3a/ : SUCCESS in 29m 43s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/19/check/gate-nova-releasenotes/57b868d//releasenotes/build/html/ : SUCCESS in 2m 51s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-nova-tox-db-functional-ubuntu-xenial/dcf7aa0/ : FAILURE in 11m 49s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/53/338553/19/check/gate-grenade-dsvm-neutron-multinode/b00f590/ : SUCCESS in 2h 21m 07s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/53/338553/19/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/8fc3521/ : SUCCESS in 37m 55s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-cells-ubuntu-xenial/cae84ac/ : SUCCESS in 49m 15s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/100db59/ : SUCCESS in 44m 53s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/32970d9/ : SUCCESS in 52m 16s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/85384df/ : SUCCESS in 1h 02m 13s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6c5df44/ : SUCCESS in 20m 52s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/19/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/59dc58d/ : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"27b91b97054a272c8a2fec264caf1df9dc0776e0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 18:54:30.000000000","message":"Patch Set 19:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/19/check/check-dsvm-tempest-vz7-exe-minimal/414a46e : FAILURE in 1h 32m 33s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"e6b93fc50f0a60b63a584ce69ab5160060cec8b0","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 19:04:00.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/19/ : SUCCESS in 2h 02m 58s","accounts_in_message":[],"_revision_number":19},{"id":"277f70d97eea578526e8729842d6fc8e46387cbb","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-11-09 19:53:17.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8404a9e3a3dababc25c590b7c5dedada50ede670","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 20:53:28.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/53/338553/20/check/dsvm-tempest-kvm/74dfb14 : SUCCESS in 59m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"1f9c7968993b0c32582d9bc2ad2bf1acc0e1ab07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-09 20:57:58.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/53/338553/20/check/gate-nova-docs-ubuntu-xenial/d60ad78//doc/build/html/ : SUCCESS in 11m 54s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-nova-pep8-ubuntu-xenial/f440c90/ : SUCCESS in 4m 24s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-nova-python27-db-ubuntu-xenial/cea8f73/ : SUCCESS in 10m 21s\n- gate-nova-python34-db http://logs.openstack.org/53/338553/20/check/gate-nova-python34-db/b4a1df1/ : SUCCESS in 7m 27s\n- gate-nova-python35-db-nv http://logs.openstack.org/53/338553/20/check/gate-nova-python35-db-nv/4fc362c/ : SUCCESS in 8m 10s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3253d48/ : SUCCESS in 1h 03m 20s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-grenade-dsvm-neutron-ubuntu-xenial/bf0823c/ : SUCCESS in 32m 24s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-grenade-dsvm-ubuntu-xenial/0659a2c/ : SUCCESS in 26m 56s\n- gate-nova-releasenotes http://docs-draft.openstack.org/53/338553/20/check/gate-nova-releasenotes/8911375//releasenotes/build/html/ : SUCCESS in 3m 34s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-nova-tox-db-functional-ubuntu-xenial/f2fe964/ : SUCCESS in 11m 17s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/53/338553/20/check/gate-grenade-dsvm-neutron-multinode/a133e05/ : SUCCESS in 38m 20s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/53/338553/20/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/b072b29/ : SUCCESS in 33m 19s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-cells-ubuntu-xenial/57c0c79/ : SUCCESS in 31m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7d732c6/ : SUCCESS in 46m 05s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/8b5e667/ : FAILURE in 53m 24s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/22590e8/ : SUCCESS in 46m 38s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d9bb8fc/ : SUCCESS in 20m 15s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/53/338553/20/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/9ca1241/ : SUCCESS in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"b1dc23890b5b268a3ad9bbf39389ebb4967d9ef5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-09 21:04:21.000000000","message":"Patch Set 20:\n\n* pci-test http://52.27.155.124/pci/338553/20 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"9e31ea6103c1bf262a56c3e8465e8afee68128d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-09 21:22:54.000000000","message":"Patch Set 20:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/20/check/tempest-dsvm-full-xenial/c7a89c2/ : FAILURE in 1h 11m 33s (non-voting)\n- tempest-dsvm-neutron-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/53/338553/20/check/tempest-dsvm-neutron-full-xenial/4815466/ : SUCCESS in 1h 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f95031409b9ed06a16a79b30398a17a16e5eba21","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 21:36:34.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/338553/20/check/check-dsvm-tempest-vz7-exe-minimal/ceb4569 : FAILURE in 1h 24m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"05ff948890243e24a91e75121a90b621142ef1c4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-09 21:43:12.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/53/338553/20/check/dsvm-tempest-xen/fc37b60 : SUCCESS in 1h 49m 25s","accounts_in_message":[],"_revision_number":20},{"id":"e03c68c33732b9720a52ef754165783e28af15a9","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-09 21:55:56.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/338553/20/ : SUCCESS in 2h 02m 09s","accounts_in_message":[],"_revision_number":20},{"id":"2e0dc125caafa490bd46fa5547d2b0f70a93f0ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-09 21:57:28.000000000","message":"Patch Set 20:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-338553-20 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":20},{"id":"d144e8a86c871dccd6dd3496339455ae2a83f5a9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-09 22:35:10.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/53/338553/20/check/dsvm-tempest-neutron-network/ab8d881 : SUCCESS in 2h 09m 56s","accounts_in_message":[],"_revision_number":20},{"id":"a2a8acfcafb605b312fce0ee86a666ac02e9c87b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-10 07:05:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/20/check/tempest-dsvm-full-nfv-xenial/3155b1a : SUCCESS in 1h 50m 50s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/53/338553/20/check/tempest-dsvm-intel-nfv-xenial/b7e302c : SUCCESS in 38m 43s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/20/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/19295fb : SUCCESS in 1h 01m 46s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/53/338553/20/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/2bb2575 : SUCCESS in 1h 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e500a668609fa7decab866890837c2a732561d5e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-10 22:54:51.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/53/338553/20/check-nova/Nova-ML2-Sriov/0b5feec : SUCCESS in 43m 13s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/53/338553/20/check-nova/Nova-MACVTAP-ML2-Sriov/5bdd28a : SUCCESS in 46m 08s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":20},{"id":"233a56336e2f56ededeb3c4ab0cb06e9a7d0c74a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-14 11:18:48.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"d77638e3ed84b349889e23763522a2e77d1d10ba","author":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"date":"2016-11-14 16:19:17.000000000","message":"Patch Set 20:\n\nUnfortunately, there is a recently found duplicate (was not attached to the bp).\n\nhttps://review.openstack.org/#/c/319532/","accounts_in_message":[],"_revision_number":20},{"id":"a83ea511b9305076c5ac7dd8d72568ead2816e96","author":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"date":"2016-11-14 18:15:13.000000000","message":"Abandoned\n\nDuplicate of https://review.openstack.org/#/c/319532.","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"2a38e06a1fcc742eb128b366f3500c4ff954963a","revisions":{"81b4e15fb6eb6855fd376ef5121f4edb36ba4cce":{"kind":"REWORK","_number":1,"created":"2016-07-06 21:04:27.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/1"}}},"commit":{"parents":[{"commit":"a0b224b17b34c4c1879362e95633f0ccf35eb502","subject":"Merge \"XenAPI: Add option for running nova independently from hypervisor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0b224b17b34c4c1879362e95633f0ccf35eb502"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-01 20:28:07.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * metadata_manager\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81b4e15fb6eb6855fd376ef5121f4edb36ba4cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81b4e15fb6eb6855fd376ef5121f4edb36ba4cce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a0b224b17b34c4c1879362e95633f0ccf35eb502","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"77062992e699cb1d64883ba7057501cea3b2315c":{"kind":"REWORK","_number":2,"created":"2016-07-11 15:20:51.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/2"}}},"commit":{"parents":[{"commit":"aca1bcebce8fdd0dc0def1382667162fb49175c8","subject":"Merge \"Body verification of os-interface.inc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aca1bcebce8fdd0dc0def1382667162fb49175c8"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-11 15:17:45.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * metadata_manager\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/77062992e699cb1d64883ba7057501cea3b2315c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/77062992e699cb1d64883ba7057501cea3b2315c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aca1bcebce8fdd0dc0def1382667162fb49175c8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"832e7b94dc7da95ec4a8af060d2113fa2a942335":{"kind":"REWORK","_number":3,"created":"2016-07-12 18:42:47.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/3"}}},"commit":{"parents":[{"commit":"aca1bcebce8fdd0dc0def1382667162fb49175c8","subject":"Merge \"Body verification of os-interface.inc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aca1bcebce8fdd0dc0def1382667162fb49175c8"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-12 17:13:44.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * metadata_manager\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/832e7b94dc7da95ec4a8af060d2113fa2a942335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/832e7b94dc7da95ec4a8af060d2113fa2a942335"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aca1bcebce8fdd0dc0def1382667162fb49175c8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"dded57225ba44748003965ed87e0ee25be2679f1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-07-13 15:07:50.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/4"}}},"commit":{"parents":[{"commit":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","subject":"Merge \"Fix PUT server tag 201 to return empty content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ebfc3dcd4734a84d7460a204d461e834d0a13ea"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-13 15:07:49.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * metadata_manager\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dded57225ba44748003965ed87e0ee25be2679f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dded57225ba44748003965ed87e0ee25be2679f1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8b16db38afd75359e139b7bce2c1aa4dcc05fd9c":{"kind":"REWORK","_number":5,"created":"2016-07-13 18:04:41.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/5"}}},"commit":{"parents":[{"commit":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","subject":"Merge \"Fix PUT server tag 201 to return empty content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ebfc3dcd4734a84d7460a204d461e834d0a13ea"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-13 18:04:23.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * metadata_manager\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b16db38afd75359e139b7bce2c1aa4dcc05fd9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b16db38afd75359e139b7bce2c1aa4dcc05fd9c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8c54a4a6a1a4108793a51b372818913bae487361":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2016-07-13 21:12:38.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/6"}}},"commit":{"parents":[{"commit":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","subject":"Merge \"Fix PUT server tag 201 to return empty content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ebfc3dcd4734a84d7460a204d461e834d0a13ea"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-13 21:12:29.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c54a4a6a1a4108793a51b372818913bae487361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c54a4a6a1a4108793a51b372818913bae487361"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"89b7552b2aed4be18775918b58fe3e33026373c0":{"kind":"REWORK","_number":7,"created":"2016-07-13 21:13:13.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/7"}}},"commit":{"parents":[{"commit":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","subject":"Merge \"Fix PUT server tag 201 to return empty content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ebfc3dcd4734a84d7460a204d461e834d0a13ea"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-13 21:13:05.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89b7552b2aed4be18775918b58fe3e33026373c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89b7552b2aed4be18775918b58fe3e33026373c0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2e9e5eabdd4da0c046ca8294ef8bcc306822c2d3":{"kind":"REWORK","_number":8,"created":"2016-07-14 17:44:16.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/8"}}},"commit":{"parents":[{"commit":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","subject":"Merge \"Fix PUT server tag 201 to return empty content\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ebfc3dcd4734a84d7460a204d461e834d0a13ea"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-07-14 17:43:52.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e9e5eabdd4da0c046ca8294ef8bcc306822c2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e9e5eabdd4da0c046ca8294ef8bcc306822c2d3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ebfc3dcd4734a84d7460a204d461e834d0a13ea","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e149a7356cdc8ba00321305feac906e3320f2494":{"kind":"REWORK","_number":9,"created":"2016-08-15 18:39:14.000000000","uploader":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"ref":"refs/changes/53/338553/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/9"}}},"commit":{"parents":[{"commit":"60ea25f52b5d68bbec8494c2113f77ac7886da13","subject":"Improve help text for service options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ea25f52b5d68bbec8494c2113f77ac7886da13"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","date":"2016-08-15 18:38:01.000000000","tz":-300},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e149a7356cdc8ba00321305feac906e3320f2494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e149a7356cdc8ba00321305feac906e3320f2494"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"60ea25f52b5d68bbec8494c2113f77ac7886da13","is_merged_in_target_branch":false,"change_id":"I543e2208b779f9ae3c5c813bba1f93e53c7ca8f4","change_number":332861,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"61c589e93082419862fc02c518a516d4c7245637":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2016-08-15 18:40:22.000000000","uploader":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"ref":"refs/changes/53/338553/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/10"}}},"commit":{"parents":[{"commit":"60ea25f52b5d68bbec8494c2113f77ac7886da13","subject":"Improve help text for service options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ea25f52b5d68bbec8494c2113f77ac7886da13"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","date":"2016-08-15 18:40:18.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-newton\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61c589e93082419862fc02c518a516d4c7245637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61c589e93082419862fc02c518a516d4c7245637"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"60ea25f52b5d68bbec8494c2113f77ac7886da13","is_merged_in_target_branch":false,"change_id":"I543e2208b779f9ae3c5c813bba1f93e53c7ca8f4","change_number":332861,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a859927f17357724953a23ee2dd3a6da0f74876c":{"kind":"REWORK","_number":11,"created":"2016-08-15 18:41:06.000000000","uploader":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"ref":"refs/changes/53/338553/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/11"}}},"commit":{"parents":[{"commit":"60ea25f52b5d68bbec8494c2113f77ac7886da13","subject":"Improve help text for service options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ea25f52b5d68bbec8494c2113f77ac7886da13"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","date":"2016-08-15 18:41:01.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-newton\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a859927f17357724953a23ee2dd3a6da0f74876c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a859927f17357724953a23ee2dd3a6da0f74876c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"60ea25f52b5d68bbec8494c2113f77ac7886da13","is_merged_in_target_branch":false,"change_id":"I543e2208b779f9ae3c5c813bba1f93e53c7ca8f4","change_number":332861,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5386a4475d2745c173b5c5fe5d1c9ca8ae8efe90":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2016-09-14 08:15:51.000000000","uploader":{"_account_id":14511,"name":"Maciej Szankin","email":"maciej+review@szankin.pl","username":"mszankin"},"ref":"refs/changes/53/338553/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/12"}}},"commit":{"parents":[{"commit":"60ea25f52b5d68bbec8494c2113f77ac7886da13","subject":"Improve help text for service options.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60ea25f52b5d68bbec8494c2113f77ac7886da13"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Maciej Szankin","email":"maciej.szankin@intel.com","date":"2016-09-14 08:15:47.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed in Newton.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5386a4475d2745c173b5c5fe5d1c9ca8ae8efe90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5386a4475d2745c173b5c5fe5d1c9ca8ae8efe90"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"60ea25f52b5d68bbec8494c2113f77ac7886da13","is_merged_in_target_branch":false,"change_id":"I543e2208b779f9ae3c5c813bba1f93e53c7ca8f4","change_number":332861,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e11b0efc7fd310c9c14eb4b0687d5d0a5844f52d":{"kind":"REWORK","_number":13,"created":"2016-09-23 22:29:40.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/13"}}},"commit":{"parents":[{"commit":"3db8995380e9e08318a1825012e0e3b309e75fc8","subject":"Merge \"Remove unused code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3db8995380e9e08318a1825012e0e3b309e75fc8"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-09-23 21:55:54.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11b0efc7fd310c9c14eb4b0687d5d0a5844f52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11b0efc7fd310c9c14eb4b0687d5d0a5844f52d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3db8995380e9e08318a1825012e0e3b309e75fc8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a7e30732d7d9604a82108b79f96281c547e681a9":{"kind":"REWORK","_number":14,"created":"2016-09-28 21:14:08.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/14"}}},"commit":{"parents":[{"commit":"3db8995380e9e08318a1825012e0e3b309e75fc8","subject":"Merge \"Remove unused code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3db8995380e9e08318a1825012e0e3b309e75fc8"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-09-28 21:12:54.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7e30732d7d9604a82108b79f96281c547e681a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7e30732d7d9604a82108b79f96281c547e681a9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3db8995380e9e08318a1825012e0e3b309e75fc8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"52f931012dbc799ac20cd59328956ba8b3a9b2ae":{"kind":"REWORK","_number":15,"created":"2016-09-29 17:44:19.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/15"}}},"commit":{"parents":[{"commit":"d5f113de34589933c23a7cecc25ca0bf5f47207d","subject":"Merge \"libvirt: ignore conflict when defining network filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5f113de34589933c23a7cecc25ca0bf5f47207d"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-09-29 17:16:44.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/52f931012dbc799ac20cd59328956ba8b3a9b2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/52f931012dbc799ac20cd59328956ba8b3a9b2ae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d5f113de34589933c23a7cecc25ca0bf5f47207d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8f294e02eb8ddc5384dbb9293234217371dd7996":{"kind":"REWORK","_number":16,"created":"2016-10-07 21:09:54.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/16"}}},"commit":{"parents":[{"commit":"d5f113de34589933c23a7cecc25ca0bf5f47207d","subject":"Merge \"libvirt: ignore conflict when defining network filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5f113de34589933c23a7cecc25ca0bf5f47207d"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-10-07 21:07:52.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f294e02eb8ddc5384dbb9293234217371dd7996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f294e02eb8ddc5384dbb9293234217371dd7996"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d5f113de34589933c23a7cecc25ca0bf5f47207d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b68c56feef502387506077647487da3f8949bb02":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2016-10-07 21:12:10.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/17"}}},"commit":{"parents":[{"commit":"d5f113de34589933c23a7cecc25ca0bf5f47207d","subject":"Merge \"libvirt: ignore conflict when defining network filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5f113de34589933c23a7cecc25ca0bf5f47207d"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-10-07 21:12:01.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nAlso added constants for service manager classes so that tests\ncan call them directly instead of remembering the class names.\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b68c56feef502387506077647487da3f8949bb02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b68c56feef502387506077647487da3f8949bb02"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d5f113de34589933c23a7cecc25ca0bf5f47207d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6842627f2c7f524b8d5ebd3fd798a4a8d658641d":{"kind":"REWORK","_number":18,"created":"2016-10-10 15:47:33.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/18"}}},"commit":{"parents":[{"commit":"d5f113de34589933c23a7cecc25ca0bf5f47207d","subject":"Merge \"libvirt: ignore conflict when defining network filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5f113de34589933c23a7cecc25ca0bf5f47207d"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-10-10 15:07:44.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nAlso added constants for service manager classes so that tests\ncan call them directly instead of remembering the class names.\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6842627f2c7f524b8d5ebd3fd798a4a8d658641d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6842627f2c7f524b8d5ebd3fd798a4a8d658641d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d5f113de34589933c23a7cecc25ca0bf5f47207d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c4a6765f2c6ddf72c57497715d8a0b490b3a857e":{"kind":"REWORK","_number":19,"created":"2016-11-09 16:24:57.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/19"}}},"commit":{"parents":[{"commit":"61b50feadeefd547f8452188f5d4a8a4219caef4","subject":"Merge \"Fix a TypeError in notification_sample_base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61b50feadeefd547f8452188f5d4a8a4219caef4"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-11-09 16:23:41.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nAlso added constants for service manager classes so that tests\ncan call them directly instead of remembering the class names.\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4a6765f2c6ddf72c57497715d8a0b490b3a857e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4a6765f2c6ddf72c57497715d8a0b490b3a857e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"61b50feadeefd547f8452188f5d4a8a4219caef4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2a38e06a1fcc742eb128b366f3500c4ff954963a":{"kind":"REWORK","_number":20,"created":"2016-11-09 19:53:17.000000000","uploader":{"_account_id":18337,"name":"Sujitha","email":"sujitha.neti@intel.com","username":"Sujitha"},"ref":"refs/changes/53/338553/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/338553/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/338553/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/338553/20"}}},"commit":{"parents":[{"commit":"61b50feadeefd547f8452188f5d4a8a4219caef4","subject":"Merge \"Fix a TypeError in notification_sample_base.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61b50feadeefd547f8452188f5d4a8a4219caef4"}]}],"author":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-06-30 22:41:02.000000000","tz":0},"committer":{"name":"Sujitha","email":"sujitha.neti@intel.com","date":"2016-11-09 19:47:21.000000000","tz":0},"subject":"Remove deprecated manager classes in service.py","message":"Remove deprecated manager classes in service.py\n\nBelow options in DEFAULT section which were deprecated in\nMitaka has been completely removed.\n      * console_manager\n      * consoleauth_manager\n      * cert_manager\n      * scheduler_manager\n\nAlso added constants for service manager classes so that tests\ncan call them directly instead of remembering the class names.\n\nBlueprint centralize-config-options-ocata\n\nChange-Id: Idc049b455a7df483556cfba0c34d5fcfb3deb3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a38e06a1fcc742eb128b366f3500c4ff954963a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a38e06a1fcc742eb128b366f3500c4ff954963a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"61b50feadeefd547f8452188f5d4a8a4219caef4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
