)]}'
{"id":"openstack%2Fnova~347893","triplet_id":"openstack%2Fnova~master~If4ea7f68c3b4be1a6d66211c4eaca8b25d2b3fca","project":"openstack/nova","branch":"master","topic":"bug/1585601","hashtags":[],"change_id":"If4ea7f68c3b4be1a6d66211c4eaca8b25d2b3fca","subject":"Refactor deallocate_fixed tests to use one mock approach instead of three","status":"MERGED","created":"2016-07-27 15:18:02.000000000","updated":"2016-08-03 12:19:23.000000000","submitted":"2016-08-03 12:19:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"82edb0f6b59966a3a10ee447878bd1e42cb125f1","_number":347893,"virtual_id_number":347893,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-08-03 12:19:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":2,"date":"2016-08-02 16:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2016-08-02 21:29:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":1,"date":"2016-08-03 08:35:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2016-08-03 10:32:01.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2016-08-03 10:32:01.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-27 23:35:20.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2016-08-01 21:37:38.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2016-08-02 02:35:02.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-02 02:55:28.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-02 03:00:28.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-02 06:37:06.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-02 16:01:12.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2016-08-02 21:29:33.000000000","updated_by":{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},"reviewer":{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},"state":"REVIEWER"},{"updated":"2016-08-03 08:35:15.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2016-08-03 10:32:01.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2016-08-03 12:19:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c474a1d6bf27be2273d66f1f8fb45d8c9fce43f6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-07-27 15:18:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8dcce9c30a85397627df0bb98e2e7874bb26f00","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-07-27 15:42:20.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/347893/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cac6321e90cbafbf4c88699cbeb1757d1734ddc5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-07-27 17:17:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/347893/1/ : SUCCESS in 1h 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"713fd5c8d6eee9fa5ed9a2e99c63ff7f08ace754","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-27 17:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/93/347893/1/check/gate-nova-docs-ubuntu-xenial/8b5320b//doc/build/html/ : SUCCESS in 9m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/93/347893/1/check/gate-nova-pep8-ubuntu-xenial/d57d49d/ : SUCCESS in 4m 34s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/93/347893/1/check/gate-nova-python27-db-ubuntu-xenial/0a78028/ : SUCCESS in 12m 50s\n- gate-nova-python34-db http://logs.openstack.org/93/347893/1/check/gate-nova-python34-db/ec0c2ed/ : SUCCESS in 13m 21s\n- gate-nova-python35-db-nv http://logs.openstack.org/93/347893/1/check/gate-nova-python35-db-nv/2019356/ : SUCCESS in 18m 00s (non-voting)\n- gate-nova-tox-db-functional http://logs.openstack.org/93/347893/1/check/gate-nova-tox-db-functional/bddf9b0/ : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9bd40bba2c451cf968799afd63867c99e188b3cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-07-27 18:10:44.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-93-347893-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"4ddcf006a7c92ce5afe67154794eff4d3eaf3283","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-07-27 18:25:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/93/347893/1/check/tempest-dsvm-full-nfv/e32422c : SUCCESS in 1h 32m 24s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/93/347893/1/check/tempest-dsvm-intel-nfv/59496ba : SUCCESS in 33m 38s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/93/347893/1/check/tempest-dsvm-ovsdpdk-nfv-networking/39663ae : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a92742ba53e76bead33d601e010a856d10688708","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-08-01 18:34:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nSeems we should use assert_called_once_with or assert_has_calls in the first test for the release_dhcp mock. Otherwise this looks good.","accounts_in_message":[],"_revision_number":1},{"id":"9dcfa59863b0ae7792614e65e6036ec457abbd19","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-01 18:47:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"34ecf617f44f9468310be15dd2154ce6b0c0d571","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-08-01 21:18:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ab4af86177c3b57ef5e14607355f889afc70a23","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-01 21:36:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d9e37a87928d454d0218ea8bab06c5e16d997bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-01 21:36:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb52c3080d11b63701a93e7ab51484c8411dc485","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-08-01 21:37:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cb81652e8c78d0bc0a9683cb225a9a8b80b221f8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-08-01 22:12:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65765797bed22b927ba972ceec199f4829e84866","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-02 00:55:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/93/347893/2/check/gate-nova-docs-ubuntu-xenial/4d16431//doc/build/html/ : SUCCESS in 13m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/93/347893/2/check/gate-nova-pep8-ubuntu-xenial/ca7032f/ : SUCCESS in 15m 02s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/93/347893/2/check/gate-nova-python27-db-ubuntu-xenial/caaeb00/ : SUCCESS in 17m 19s\n- gate-nova-python34-db http://logs.openstack.org/93/347893/2/check/gate-nova-python34-db/b4d9ebf/ : SUCCESS in 13m 58s\n- gate-nova-python35-db-nv http://logs.openstack.org/93/347893/2/check/gate-nova-python35-db-nv/3855fb7/ : SUCCESS in 19m 08s (non-voting)\n- gate-nova-tox-db-functional http://logs.openstack.org/93/347893/2/check/gate-nova-tox-db-functional/510b7fc/ : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":2},{"id":"72841219ac53df5976822d7417481a59f1aefbfd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-02 02:35:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/93/347893/2/check/tempest-dsvm-full-nfv/43991d1 : SUCCESS in 1h 28m 45s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/93/347893/2/check/tempest-dsvm-intel-nfv/3badb74 : SUCCESS in 33m 08s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/93/347893/2/check/tempest-dsvm-ovsdpdk-nfv-networking/dc1ef07 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"103464ef12b161009b5ffaa5e152fd0abd8de5f9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-02 02:55:28.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/347893/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"78f6220c80660af3e843be6be8408d48376d4996","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-08-02 03:00:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/347893/2/ : SUCCESS in 1h 55m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ef72c7ae29756aed76370374023f3c224e4012a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-08-02 06:37:06.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-93-347893-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"0713de1c8450bfea1a9610f388dceca7db5a4942","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-02 13:50:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5229aeb06283479e71e81c0f61dbeac2331fe61d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-08-02 15:17:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33e38a17b7d68721d8fa6faf9991127ead4a4e1d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-02 15:20:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aacf8a9a63597fd0f4ae56ec743f252103f320c3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-02 15:55:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27d852ea78defaabdf4d330199575f477c5cef7d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2016-08-02 15:56:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"47e1ca9815722c7fedbf0dbabb29b0be61e9e4a8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2016-08-02 16:01:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f0c8428ef80ac78aa60403c9446defeefe725f98","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-02 17:10:12.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/347893/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c2be17b1d8a2f41d50946cd1bebf2c5ed64a3c4e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-02 18:08:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/93/347893/3/check/tempest-dsvm-full-nfv/e67963d : SUCCESS in 1h 26m 57s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/93/347893/3/check/tempest-dsvm-intel-nfv/2391a12 : SUCCESS in 40m 06s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/93/347893/3/check/tempest-dsvm-ovsdpdk-nfv-networking/619d045 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":3},{"id":"effd4dab9661aff5137d53bc9ee9b35c316e41af","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-08-02 18:49:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/347893/3/ : SUCCESS in 1h 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"715c033aa2dc21b29572cfe10c17497d139882a5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-02 20:12:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/93/347893/3/check/gate-nova-docs-ubuntu-xenial/928c929//doc/build/html/ : SUCCESS in 8m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/93/347893/3/check/gate-nova-pep8-ubuntu-xenial/79ae5a2/ : SUCCESS in 5m 09s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/93/347893/3/check/gate-nova-python27-db-ubuntu-xenial/b8da706/ : SUCCESS in 17m 24s\n- gate-nova-python34-db http://logs.openstack.org/93/347893/3/check/gate-nova-python34-db/305581d/ : SUCCESS in 11m 56s\n- gate-nova-python35-db-nv http://logs.openstack.org/93/347893/3/check/gate-nova-python35-db-nv/39426a9/ : SUCCESS in 17m 52s (non-voting)\n- gate-nova-tox-db-functional http://logs.openstack.org/93/347893/3/check/gate-nova-tox-db-functional/48157e2/ : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f9de967e44bf809c564dbe80790d86f002382e06","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-08-02 21:01:58.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-93-347893-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"71f340d2bdee32a2f1d91936565b66201fdef887","author":{"_account_id":2506,"name":"devdatta-kulkarni","email":"kulkarni.devdatta@gmail.com","username":"devdatta-kulkarni"},"date":"2016-08-02 21:29:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"10df5a7c5835dcc0989da54dabec07f7d62d678a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2016-08-03 08:35:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2e590ae4adc806911f517794f07f099409317bfd","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2016-08-03 10:32:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"55f8abb8bb6b9d94c9c334c6972ffb5dffb98459","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-03 10:32:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"7d06da5dc9f4b432fcd4cf78d26568bb0a9246cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-03 12:19:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/93/347893/3/gate/gate-nova-docs-ubuntu-xenial/f954d81//doc/build/html/ : SUCCESS in 8m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/93/347893/3/gate/gate-nova-pep8-ubuntu-xenial/66a58d0/ : SUCCESS in 5m 32s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/93/347893/3/gate/gate-nova-python27-db-ubuntu-xenial/a6f0869/ : SUCCESS in 12m 27s\n- gate-nova-python34-db http://logs.openstack.org/93/347893/3/gate/gate-nova-python34-db/e5d651d/ : SUCCESS in 13m 27s\n- gate-nova-tox-db-functional http://logs.openstack.org/93/347893/3/gate/gate-nova-tox-db-functional/4b71be7/ : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ed5bf2e750141ea0fc8974826bb25df9e0252e41","date":"2016-08-03 12:19:23.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"00ebb729825de302fcf043edfb374f4cc5367869","revisions":{"f7fea37c7c72f97b6fe12b10a459806648f195fc":{"kind":"REWORK","_number":1,"created":"2016-07-27 15:18:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/93/347893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/93/347893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/93/347893/1"}}},"commit":{"parents":[{"commit":"f9e9b30b93443c986d3bee7f8b4140b82050418b","subject":"Call release_dhcp via RPC to ensure correct host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9e9b30b93443c986d3bee7f8b4140b82050418b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-07-27 15:20:20.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-07-27 15:20:20.000000000","tz":-420},"subject":"Refactor deallocate_fixed tests to use one mock approach instead of three","message":"Refactor deallocate_fixed tests to use one mock approach instead of three\n\nThis cleans up the tests added/edited in the last patch per feedback.\nInstead of mock, stubs, and mox, this just uses mock.\n\nChange-Id: If4ea7f68c3b4be1a6d66211c4eaca8b25d2b3fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7fea37c7c72f97b6fe12b10a459806648f195fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7fea37c7c72f97b6fe12b10a459806648f195fc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9e9b30b93443c986d3bee7f8b4140b82050418b","is_merged_in_target_branch":true,"change_id":"I0eec8c995dd8cff50c37af83018697fc686fe727","change_number":325361,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c9dc0f831991b7e20af675806883b2135bd3883a":{"kind":"REWORK","_number":2,"created":"2016-08-01 21:36:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/93/347893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/93/347893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/93/347893/2"}}},"commit":{"parents":[{"commit":"f9e9b30b93443c986d3bee7f8b4140b82050418b","subject":"Call release_dhcp via RPC to ensure correct host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9e9b30b93443c986d3bee7f8b4140b82050418b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-07-27 15:20:20.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-08-01 21:39:36.000000000","tz":-420},"subject":"Refactor deallocate_fixed tests to use one mock approach instead of three","message":"Refactor deallocate_fixed tests to use one mock approach instead of three\n\nThis cleans up the tests added/edited in the last patch per feedback.\nInstead of mock, stubs, and mox, this just uses mock.\n\nChange-Id: If4ea7f68c3b4be1a6d66211c4eaca8b25d2b3fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c9dc0f831991b7e20af675806883b2135bd3883a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c9dc0f831991b7e20af675806883b2135bd3883a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9e9b30b93443c986d3bee7f8b4140b82050418b","is_merged_in_target_branch":true,"change_id":"I0eec8c995dd8cff50c37af83018697fc686fe727","change_number":325361,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"00ebb729825de302fcf043edfb374f4cc5367869":{"kind":"REWORK","_number":3,"created":"2016-08-02 15:56:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/93/347893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/93/347893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/93/347893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/93/347893/3"}}},"commit":{"parents":[{"commit":"f9e9b30b93443c986d3bee7f8b4140b82050418b","subject":"Call release_dhcp via RPC to ensure correct host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9e9b30b93443c986d3bee7f8b4140b82050418b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-07-27 15:20:20.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2016-08-02 15:59:44.000000000","tz":-420},"subject":"Refactor deallocate_fixed tests to use one mock approach instead of three","message":"Refactor deallocate_fixed tests to use one mock approach instead of three\n\nThis cleans up the tests added/edited in the last patch per feedback.\nInstead of mock, stubs, and mox, this just uses mock.\n\nChange-Id: If4ea7f68c3b4be1a6d66211c4eaca8b25d2b3fca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00ebb729825de302fcf043edfb374f4cc5367869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00ebb729825de302fcf043edfb374f4cc5367869"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f9e9b30b93443c986d3bee7f8b4140b82050418b","is_merged_in_target_branch":true,"change_id":"I0eec8c995dd8cff50c37af83018697fc686fe727","change_number":325361,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
