)]}'
{"id":"openstack%2Fnova~352556","triplet_id":"openstack%2Fnova~master~Ib9c08e4f9becf9500599d20e8f93ac75f2a77b5f","project":"openstack/nova","branch":"master","topic":"bug/1611008","hashtags":[],"change_id":"Ib9c08e4f9becf9500599d20e8f93ac75f2a77b5f","subject":"Add some logging and a comment for shelve/unshelve operations","status":"MERGED","created":"2016-08-08 18:49:51.000000000","updated":"2016-08-09 23:34:35.000000000","submitted":"2016-08-09 10:22:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0b2edb9a96806789c19984be67fb4b4c4e87308","_number":352556,"virtual_id_number":352556,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"recommended":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":1,"date":"2016-08-08 20:17:40.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-08-09 10:22:45.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2016-08-09 23:34:35.000000000","post_submit":true,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":1,"date":"2016-08-08 21:23:14.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"all":[{"value":2,"date":"2016-08-08 18:56:36.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-08-09 06:31:00.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-08-09 06:31:00.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-08-08 18:56:36.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2016-08-08 20:17:40.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-08 21:23:14.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-08-09 06:31:00.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2016-08-09 10:22:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-08-09 23:34:35.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1310d28d92cae8aa2a67ee91b4549cef64613df7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2016-08-08 18:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a40f68f2c8c74e9cdc36fd37599be2bd80293c8","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-08-08 18:56:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e2b40f41f5da5ed454bb32c8fae3adcd4bdfb9f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-08-08 19:44:45.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/352556/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3f595a2572aaa611114c8e81a2eacf7014c408cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-08-08 20:00:08.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/56/352556/1/check/check-ibm-tempest-dsvm-full/5b7fc67/ : SUCCESS in 1h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d163c34adec11dc909a11769aa1cb69fdec2cce5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-08-08 20:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/56/352556/1/check/dsvm-tempest-nova-network/c65e0a5 : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ab9829d27341dfb21ce9ed8f16c1caf1c20d4826","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-08-08 21:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv http://intel-openstack-ci-logs.ovh/56/352556/1/check/tempest-dsvm-full-nfv/f796254 : SUCCESS in 1h 29m 31s\n- tempest-dsvm-intel-nfv http://intel-openstack-ci-logs.ovh/56/352556/1/check/tempest-dsvm-intel-nfv/ddd1db3 : SUCCESS in 34m 08s\n- tempest-dsvm-ovsdpdk-nfv-networking http://intel-openstack-ci-logs.ovh/56/352556/1/check/tempest-dsvm-ovsdpdk-nfv-networking/4131374 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f8f65d0636e6d916a2e2c976483533800668457a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-08-08 22:21:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/352556/1/ : FAILURE in 3h 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7afc9122221c948092b195d494e8285371f1813","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-08-08 22:55:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/56/352556/1/check-nova/Nova-ML2-Sriov/4f0d639 : SUCCESS in 32m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/56/352556/1/check-nova/Nova-MACVTAP-ML2-Sriov/c5812b6 : SUCCESS in 31m 46s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"4078cc0bb15b1a39901e39f8f4421c281f7757ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 23:17:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/352556/1/check/gate-nova-docs-ubuntu-xenial/1cb1a68//doc/build/html/ : SUCCESS in 8m 05s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/352556/1/check/gate-nova-pep8-ubuntu-xenial/2d106fa/ : SUCCESS in 9m 47s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/56/352556/1/check/gate-nova-python27-db-ubuntu-xenial/b0f872f/ : SUCCESS in 11m 49s\n- gate-nova-python34-db http://logs.openstack.org/56/352556/1/check/gate-nova-python34-db/d082ece/ : SUCCESS in 12m 23s\n- gate-nova-python35-db-nv http://logs.openstack.org/56/352556/1/check/gate-nova-python35-db-nv/f743785/ : SUCCESS in 15m 53s (non-voting)\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-full-ubuntu-xenial/d45bd2d/ : SUCCESS in 59m 47s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-postgres-full-ubuntu-xenial/1f4e9a3/ : SUCCESS in 48m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/34a2563/ : SUCCESS in 1h 27m 52s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/56/352556/1/check/gate-grenade-dsvm-ubuntu-trusty/d4cfe2b/ : SUCCESS in 46m 48s\n- gate-nova-tox-db-functional http://logs.openstack.org/56/352556/1/check/gate-nova-tox-db-functional/0bf52a2/ : SUCCESS in 16m 05s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/56/352556/1/check/gate-grenade-dsvm-multinode/bb39fc0/ : SUCCESS in 59m 25s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa/340b9b7/ : SUCCESS in 31m 43s\n- gate-tempest-dsvm-cells http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-cells/7e03965/ : SUCCESS in 41m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph/9eb7e3d/ : SUCCESS in 1h 08m 54s\n- gate-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-neutron-multinode-full/c45ca90/ : SUCCESS in 1h 29m 03s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration http://logs.openstack.org/56/352556/1/check/gate-tempest-dsvm-multinode-live-migration/807aae2/ : SUCCESS in 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6229944e240c7a5fb2796372f267adb031b14aec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-08-08 23:56:54.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-56-352556-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"b0301afd46d5ad44fc15bc91da302fb43d781eb3","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 01:22:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/56/352556/1/check/dsvm-tempest-kvm/bacbd98 : FAILURE in 1h 19m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"678e9c95d1d0d18808bdf0a67afd58bbcaa57650","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-08-09 04:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/56/352556/1/check/check-dsvm-tempest-vz7-exe-minimal/4567442 : SUCCESS in 2h 20m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"f56ba217f2c0f5576efdd27935c9d98ac8e2644d","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2016-08-09 06:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5769515c30e2abc5737d6ea3342680c7ad79d7a4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 06:31:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"58896cc05bc271e69a356930b2527f28305320e9","date":"2016-08-09 10:22:45.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"6559f4cf507937c6d002952d55d83a3c3600c943","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 10:22:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/352556/1/gate/gate-nova-docs-ubuntu-xenial/115059c//doc/build/html/ : SUCCESS in 5m 03s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/352556/1/gate/gate-nova-pep8-ubuntu-xenial/51549b6/ : SUCCESS in 7m 51s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/56/352556/1/gate/gate-nova-python27-db-ubuntu-xenial/aea88c3/ : SUCCESS in 12m 02s\n- gate-nova-python34-db http://logs.openstack.org/56/352556/1/gate/gate-nova-python34-db/c003da0/ : SUCCESS in 10m 06s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/gate/gate-tempest-dsvm-full-ubuntu-xenial/ce09454/ : SUCCESS in 48m 49s\n- gate-tempest-dsvm-postgres-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/gate/gate-tempest-dsvm-postgres-full-ubuntu-xenial/ed13df2/ : SUCCESS in 45m 10s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/56/352556/1/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c30813c/ : SUCCESS in 1h 15m 31s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/56/352556/1/gate/gate-grenade-dsvm-ubuntu-trusty/3cdb797/ : SUCCESS in 35m 23s\n- gate-nova-tox-db-functional http://logs.openstack.org/56/352556/1/gate/gate-nova-tox-db-functional/3838aac/ : SUCCESS in 15m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/56/352556/1/gate/gate-grenade-dsvm-multinode/62c0021/ : SUCCESS in 39m 55s\n- gate-tempest-dsvm-cells http://logs.openstack.org/56/352556/1/gate/gate-tempest-dsvm-cells/c951c73/ : SUCCESS in 31m 00s\n- gate-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/56/352556/1/gate/gate-tempest-dsvm-full-devstack-plugin-ceph/dfdeb82/ : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"200d83093246b3f666c8f77feec0fd0ecf94e4d3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-08-09 23:34:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/56/352556/1/check/dsvm-tempest-xen/c76434f : SUCCESS in 1h 31m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1bc4696b3ea6c5317a82e5e5d77dbf1ca0eb923","revisions":{"c1bc4696b3ea6c5317a82e5e5d77dbf1ca0eb923":{"kind":"REWORK","_number":1,"created":"2016-08-08 18:49:51.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/56/352556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/352556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/352556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/352556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/352556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/352556/1"}}},"commit":{"parents":[{"commit":"e285eb1a382e6d3ce1cc596eeb5cecb3b165a228","subject":"Run shelve/shelve_offload_instance in a semaphore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e285eb1a382e6d3ce1cc596eeb5cecb3b165a228"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-08-08 18:47:56.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2016-08-08 18:47:56.000000000","tz":-240},"subject":"Add some logging and a comment for shelve/unshelve operations","message":"Add some logging and a comment for shelve/unshelve operations\n\nSince shelve_instance calls shelve_offload_instance it would\nbe useful to have some logging when we start those operations.\nDo the same for unshelve since that is currently being called\nduring a Tempest run while shelve_offload_instance is still\nrunning.\n\nAlso add a comment about why we have to set the instance.vm_state\nbefore updating the resource tracker in shelve_offload_instance.\n\nChange-Id: Ib9c08e4f9becf9500599d20e8f93ac75f2a77b5f\nRelated-Bug: #1611008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1bc4696b3ea6c5317a82e5e5d77dbf1ca0eb923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1bc4696b3ea6c5317a82e5e5d77dbf1ca0eb923"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e285eb1a382e6d3ce1cc596eeb5cecb3b165a228","is_merged_in_target_branch":true,"change_id":"Id36b3b9516d72d28519c18c38d98b646b47d288d","change_number":352554,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
