)]}'
{"id":"openstack%2Fnova~369180","triplet_id":"openstack%2Fnova~master~If1cd6f3a635759cd55d116a34ca164630c61e085","project":"openstack/nova","branch":"master","topic":"bug/1622854","hashtags":[],"change_id":"If1cd6f3a635759cd55d116a34ca164630c61e085","subject":"pci: remove pci device from claims and allocations when freeing it","status":"MERGED","created":"2016-09-13 07:22:28.000000000","updated":"2016-11-04 03:10:45.000000000","submitted":"2016-11-03 17:19:54.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"832045cfa0a1aac2fe6db06d8c1869d1b19c675a","_number":369180,"virtual_id_number":369180,"owner":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-11-03 17:19:54.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":1,"date":"2016-11-03 15:38:28.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2016-11-03 18:42:36.000000000","post_submit":true,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},{"value":0,"date":"2016-11-04 03:10:45.000000000","post_submit":true,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"all":[{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2016-11-03 14:21:02.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"all":[{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-11-03 14:21:02.000000000","_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},{"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-15 14:53:42.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-09-18 07:39:11.000000000","updated_by":{"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},"reviewer":{"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},"state":"REVIEWER"},{"updated":"2016-09-19 11:57:40.000000000","updated_by":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"reviewer":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"state":"REVIEWER"},{"updated":"2016-10-09 03:57:56.000000000","updated_by":{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},"reviewer":{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},"state":"REVIEWER"},{"updated":"2016-11-02 08:59:33.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 09:12:47.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 09:20:57.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 09:26:48.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 09:56:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 09:58:06.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 10:47:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-02 10:49:52.000000000","updated_by":{"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},"reviewer":{"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},"state":"REVIEWER"},{"updated":"2016-11-02 12:01:20.000000000","updated_by":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"reviewer":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"state":"REVIEWER"},{"updated":"2016-11-02 18:53:47.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2016-11-02 19:27:25.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-03 14:21:02.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2016-11-03 15:38:28.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-03 17:19:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-11-03 18:42:36.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-04 03:10:45.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08673f5c0d31a0612a383dc506fc592da2916828","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-13 07:22:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cda41d3a6f7492e33f15dc662eb2580ea015400","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-13 07:48:30.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/369180/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8abee8ffdc4fe980ed3fc5bde314cb5dee98cc68","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-13 08:23:29.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/1/check/check-ibm-tempest-dsvm-full/334ef22/ : SUCCESS in 1h 00m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"935e366a29a845e60c14632ad2faa0908d94bf28","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-13 08:36:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/1/check/gate-nova-docs-ubuntu-xenial/df644f8//doc/build/html/ : SUCCESS in 3m 50s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-nova-pep8-ubuntu-xenial/4b8edb8/ : SUCCESS in 6m 34s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-nova-python27-db-ubuntu-xenial/feea0c8/ : SUCCESS in 14m 56s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/1/check/gate-nova-python34-db/e174a14/ : SUCCESS in 12m 05s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/1/check/gate-nova-python35-db-nv/c1d968f/ : SUCCESS in 12m 52s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/1/check/gate-grenade-dsvm-neutron-ubuntu-trusty/81e16e0/ : SUCCESS in 42m 16s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/2c22d16/ : SUCCESS in 1h 01m 05s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-full-ubuntu-xenial/7edc89b/ : SUCCESS in 48m 09s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/1/check/gate-grenade-dsvm-ubuntu-trusty/991904c/ : SUCCESS in 35m 51s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-nova-tox-db-functional-ubuntu-xenial/bfebe7a/ : SUCCESS in 13m 49s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/1/check/gate-grenade-dsvm-multinode/fa7b076/ : SUCCESS in 43m 31s\n- gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-ironic-ipa-partition-pxe_ssh-tinyipa-nv/884b5ec/ : SUCCESS in 29m 23s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/2036232/ : SUCCESS in 44m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/bb84b5f/ : SUCCESS in 49m 28s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/d11742b/ : SUCCESS in 1h 08m 00s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/4cac0ef/ : SUCCESS in 1h 03m 57s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/7f23349/ : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f5d4ebe66792754a2e5335cc20c0524f8d509b30","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-13 08:38:35.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-369180-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"4b22268de25cae52778dd23d64be2f30abd61600","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-13 08:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/1/check/dsvm-tempest-xen/094628a : SUCCESS in 1h 34m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e063d8ea44b92b5853db1c187b03026d5ed77bd1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-13 09:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/1/check-nova/Nova-ML2-Sriov/6b6ffed : SUCCESS in 32m 39s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/1/check-nova/Nova-MACVTAP-ML2-Sriov/68c5a4c : SUCCESS in 33m 53s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"09eb67409d37ebe4c80df45b239e689b2a4a1e31","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-13 09:24:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/1/ : SUCCESS in 2h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"941ed158116e83ad273c5ed0c8d9a29983928b3f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-13 09:28:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/1/check/dsvm-tempest-nova-network/dc8f843 : SUCCESS in 1h 47m 56s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/1/check/dsvm-tempest-neutron-network/e18577f : SUCCESS in 1h 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e45efe37b58b6dbe6253e0c2ae21823bf95e7b18","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-13 10:02:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/1/check/tempest-dsvm-full-nfv-xenial/17713cf : FAILURE in 2h 16m 16s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/1/check/tempest-dsvm-intel-nfv-xenial/91db0bc : SUCCESS in 44m 57s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/02ef864 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9c33b3cb74830227cc104cfa0f39f82cdf80e976","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-13 10:08:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/80/369180/1/check/dsvm-tempest-kvm/ff870a2 : SUCCESS in 1h 12m 17s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"d8eaee7d513228914c7a25ded2c841c61bbff3a3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-13 10:19:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/369180/1/check/check-dsvm-tempest-vz7-exe-minimal/44d1bab : SUCCESS in 1h 35m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"7fe289189c1bc61442915df220e8cfa9e8be9bd6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-13 11:11:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/369180/1/966/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2d3bfd67053ced3378d53bd9ff19120854790a4d","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-09-13 12:12:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"65ffba668e5fb4d6ec39b2844c733435932127d6","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-13 14:36:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"cb4f9a5427955a351654bd71e480e4b23fda4e38","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-15 11:38:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b9f7635c4437c00b0681935bdba45e3efac7f53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-15 11:43:09.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/2/check/check-ibm-tempest-dsvm-full/3a204d9/ : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf35755893b5b9cd92852782f1f1e43a077d616c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 12:29:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/2/ : FAILURE in 50m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e1f423b5920a5304f19804124916dcf7ab38a5c4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 13:12:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/369180/2/check/check-dsvm-tempest-vz7-exe-minimal/33ae007 : SUCCESS in 1h 17m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027virtuozzo-recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"db5e46fc75bfd7fd8d3058615140e519ef934eb0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-15 13:13:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/2/check/gate-nova-docs-ubuntu-xenial/6f0ddcc//doc/build/html/ : SUCCESS in 3m 49s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-pep8-ubuntu-xenial/4a4525c/ : SUCCESS in 4m 25s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-python27-db-ubuntu-xenial/36326ca/ : SUCCESS in 13m 08s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/2/check/gate-nova-python34-db/565abcb/ : SUCCESS in 10m 30s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/2/check/gate-nova-python35-db-nv/479d855/ : SUCCESS in 12m 46s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-neutron-ubuntu-trusty/2fef43f/ : SUCCESS in 1h 11m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8af98fb/ : FAILURE in 1h 08m 45s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-ubuntu-xenial/4700323/ : SUCCESS in 52m 04s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-ubuntu-trusty/84c4a28/ : SUCCESS in 44m 00s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-tox-db-functional-ubuntu-xenial/c5b40eb/ : SUCCESS in 15m 37s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-multinode/4b816ee/ : SUCCESS in 42m 41s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/5897c38/ : SUCCESS in 39m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/cd02567/ : SUCCESS in 47m 10s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/3b5e9e2/ : SUCCESS in 1h 31m 34s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/2e90b14/ : SUCCESS in 1h 09m 03s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/066cf5e/ : SUCCESS in 30m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/0032be2/ : SUCCESS in 26m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36111e7406f77102e790bb946784c9ad7b190d20","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-15 13:14:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"388a13a76107f252508329890e1c4284c3de0014","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-15 13:28:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/2/check/dsvm-tempest-nova-network/f712b68 : SUCCESS in 1h 31m 58s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/2/check/dsvm-tempest-neutron-network/1ab229e : SUCCESS in 1h 48m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc6af72ca7eef351229469aa7ef6c62800e2ec7a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 13:48:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/2/ : FAILURE in 33m 48s","accounts_in_message":[],"_revision_number":2},{"id":"58c71a858fe31ee2c917811720fb5df57e62e122","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-15 14:30:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/2/check/dsvm-tempest-xen/2d53825 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b7846af7949d12f511b4761d5d3f0b71bc4585c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-15 14:30:16.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/2/check/check-ibm-tempest-dsvm-full/bb6d78a/ : SUCCESS in 1h 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d08f2e18ad3fc39e1e682a839df938e6a6a7755","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-15 14:38:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/2/check/gate-nova-docs-ubuntu-xenial/d4d7ec3//doc/build/html/ : SUCCESS in 2m 50s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-pep8-ubuntu-xenial/101b2aa/ : SUCCESS in 4m 53s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-python27-db-ubuntu-xenial/d15bc3a/ : SUCCESS in 15m 01s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/2/check/gate-nova-python34-db/83d9d12/ : SUCCESS in 11m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/2/check/gate-nova-python35-db-nv/095bf08/ : SUCCESS in 13m 30s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-neutron-ubuntu-trusty/7ed94c6/ : FAILURE in 49m 04s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a01c775/ : SUCCESS in 1h 12m 28s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-ubuntu-xenial/0d266e8/ : SUCCESS in 58m 38s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-ubuntu-trusty/ef2a1b7/ : SUCCESS in 25m 26s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-tox-db-functional-ubuntu-xenial/33d4558/ : SUCCESS in 18m 42s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-multinode/49f749d/ : SUCCESS in 39m 27s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/59e26fd/ : SUCCESS in 38m 17s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/56eae64/ : SUCCESS in 58m 57s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/1e8640f/ : SUCCESS in 1h 18m 18s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/74a54d6/ : SUCCESS in 1h 00m 01s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a89392d/ : SUCCESS in 18m 08s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/e117a69/ : SUCCESS in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c895eb44e2d8ea9426a58d769aec1dc447362286","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 14:51:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/80/369180/2/check/dsvm-tempest-kvm/8a44386 : FAILURE in 1h 46m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"7fe9e48a8931c1480ed3ba73d999b4fc4de7a966","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-15 14:53:42.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-neutron-ubuntu-trusty: https://bugs.launchpad.net/bugs/1323658\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"3ee53f56753ec1966b23612e3b9377faad14666f","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-15 15:13:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"63e4d0f022533d7b26d0c710bff4bbfa8ef13297","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 15:19:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/2/check/tempest-dsvm-full-nfv-xenial/272054f : FAILURE in 2h 06m 55s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/2/check/tempest-dsvm-intel-nfv-xenial/56f8be3 : SUCCESS in 47m 41s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5474c20 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9e9a19d96a693e978adc01f0ace0aa48eaefa58b","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-09-15 15:29:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21c8b9f2db57d38ffb30ada98510f721622f5c5b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-15 15:46:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/2/ : FAILURE in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"64c522c04c5917da7577e30244469d8d815f6ca5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-15 16:12:45.000000000","message":"Patch Set 2:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/2/check/check-ibm-tempest-dsvm-full/7c70d62/ : SUCCESS in 58m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f315c9779af78174d336b64e562c3eea4b9998f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-15 16:38:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/2/check-nova/Nova-ML2-Sriov/1a7d8f6 : SUCCESS in 36m 09s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/2/check-nova/Nova-MACVTAP-ML2-Sriov/d199f60 : SUCCESS in 36m 38s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"caeab246aaf3c84b23d7ee604472c07a64571db5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-15 16:45:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/2/check/gate-nova-docs-ubuntu-xenial/a3d99fa//doc/build/html/ : SUCCESS in 6m 27s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-pep8-ubuntu-xenial/0432d2f/ : SUCCESS in 8m 54s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-python27-db-ubuntu-xenial/5821d59/ : SUCCESS in 15m 59s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/2/check/gate-nova-python34-db/3741d2b/ : SUCCESS in 14m 39s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/2/check/gate-nova-python35-db-nv/d5ba3b2/ : SUCCESS in 13m 20s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-neutron-ubuntu-trusty/c9688ca/ : SUCCESS in 49m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/1ab1a9f/ : SUCCESS in 1h 19m 53s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-ubuntu-xenial/8fa616e/ : SUCCESS in 51m 20s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-ubuntu-trusty/3f8ea12/ : SUCCESS in 37m 44s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-nova-tox-db-functional-ubuntu-xenial/5addc6f/ : SUCCESS in 21m 28s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/2/check/gate-grenade-dsvm-multinode/1acacb3/ : SUCCESS in 29m 47s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/192a98a/ : SUCCESS in 39m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7ca3a40/ : SUCCESS in 1h 04m 49s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/56692ac/ : SUCCESS in 1h 16m 30s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f5b4ef7/ : SUCCESS in 1h 08m 57s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3d3c2f2/ : SUCCESS in 30m 44s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/d42044e/ : SUCCESS in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"527cbc2ec4eba063a932248858472effacf78ba9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-15 18:44:58.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-369180-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"6602b0162179dc9edcfbbf0bcdfd98b37cd56d2e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-16 00:03:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/369180/2/1092/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4e441eaa1a0fc81480e4320f11c06aa11b0b7092","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-18 05:50:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e287c3e5d1d3ad00e9657a6933169b7669ead56","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-18 06:26:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/3/check-nova/Nova-ML2-Sriov/a35456b : SUCCESS in 32m 37s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/3/check-nova/Nova-MACVTAP-ML2-Sriov/0e0a4a5 : SUCCESS in 33m 24s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"d36b1c0269b3388666a5793b443f21517cd695fd","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-18 06:27:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"60a9a2eeedbf5834234625d1b44764e1231c98fd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-18 06:27:24.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/369180/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3c52ff773299b6a8babff6a2ff0eec77ef8a9b5c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-18 06:50:30.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/3/check/check-ibm-tempest-dsvm-full/d7e3c51/ : SUCCESS in 59m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4a41bb5fc80565fb635f1fe6a2132666f37cb319","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-18 07:20:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/3/check/gate-nova-docs-ubuntu-xenial/d4bd5f9//doc/build/html/ : SUCCESS in 4m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-nova-pep8-ubuntu-xenial/d06e435/ : SUCCESS in 12m 30s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-nova-python27-db-ubuntu-xenial/78b238f/ : SUCCESS in 14m 11s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/3/check/gate-nova-python34-db/cf616fb/ : SUCCESS in 10m 39s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/3/check/gate-nova-python35-db-nv/f774627/ : SUCCESS in 12m 34s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/3/check/gate-grenade-dsvm-neutron-ubuntu-trusty/7d34937/ : SUCCESS in 43m 50s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8d40a7a/ : SUCCESS in 52m 29s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-full-ubuntu-xenial/35f23a3/ : SUCCESS in 48m 11s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/3/check/gate-grenade-dsvm-ubuntu-trusty/120cac8/ : SUCCESS in 36m 14s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-nova-tox-db-functional-ubuntu-xenial/9fa7edd/ : SUCCESS in 15m 11s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/3/check/gate-grenade-dsvm-multinode/6b14d4f/ : SUCCESS in 40m 22s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/c1f4d2d/ : SUCCESS in 37m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/2776b9d/ : SUCCESS in 1h 10m 29s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/da2d762/ : FAILURE in 1h 29m 20s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/07323ae/ : SUCCESS in 1h 01m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b2a7930/ : SUCCESS in 28m 59s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/38543b5/ : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5efb9f11c3dd4b59a61da6e800951aef7895baa0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-18 07:31:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/3/check/dsvm-tempest-xen/aba5673 : SUCCESS in 1h 40m 40s","accounts_in_message":[],"_revision_number":3},{"id":"1529fe23524b3f36d2e79aeb90995d548a61db0d","author":{"_account_id":21683,"name":"edan david","email":"edand@mellanox.com","username":"edand"},"date":"2016-09-18 07:39:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1fefdf94377d4c5a37049451ddda18ba53712bb2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-18 07:42:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/3/check/tempest-dsvm-full-nfv-xenial/5724ea6 : SUCCESS in 1h 51m 13s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/3/check/tempest-dsvm-intel-nfv-xenial/27b9370 : SUCCESS in 43m 20s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/79785c2 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c57f6af925fbeb029a7381dd55b5468f48ea0bb4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-18 07:42:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/3/check/dsvm-tempest-nova-network/3da5df7 : SUCCESS in 1h 26m 08s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/3/check/dsvm-tempest-neutron-network/ed67b15 : SUCCESS in 1h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"117235a91eb1bcb185a6ef905b6761211b156961","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-18 07:55:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/3/ : SUCCESS in 2h 04m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3867d622552a7931c22d76724f3c391dbda9ff72","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-18 11:23:23.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-369180-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"d793a0b49ae99100c0028997cabda8de9e3fbd18","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-09-18 14:03:50.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"672f18f601eff353ae6cd78f695b3596576bdf57","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-18 14:25:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d80aae418d2d7594b47781f8b79da0250f46e6c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-18 14:26:30.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"23514b2213685d891c6863ea2e68e53d457c0dad","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-09-18 14:27:54.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"297f41a49150baecf8ccb0f9cd4c41317f2f89c2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-18 14:40:58.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/369180/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1207bdedd8813706fb1fe5f26d7acc9f3a85a59e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-18 14:49:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/4/check-nova/Nova-ML2-Sriov/b232c1c : FAILURE in 20m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/4/check-nova/Nova-MACVTAP-ML2-Sriov/b558aac : FAILURE in 21m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"18c6025dd4cce3fee36f14addf5ce1b58847d8e0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-18 14:53:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/4/check/dsvm-tempest-xen/730e066 : FAILURE in 26m 52s","accounts_in_message":[],"_revision_number":4},{"id":"44f6982f2cad4e28d83b38d55da724cc419ab08e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-18 15:05:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/4/check/gate-nova-docs-ubuntu-xenial/a0f42e9//doc/build/html/ : SUCCESS in 3m 38s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-nova-pep8-ubuntu-xenial/c4a7b48/ : FAILURE in 6m 47s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-nova-python27-db-ubuntu-xenial/cc8565b/ : FAILURE in 4m 17s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/4/check/gate-nova-python34-db/2e283f8/ : FAILURE in 2m 47s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/4/check/gate-nova-python35-db-nv/30ba73f/ : FAILURE in 5m 19s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/4/check/gate-grenade-dsvm-neutron-ubuntu-trusty/0a633ef/ : FAILURE in 32m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a83b691/ : FAILURE in 22m 25s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-full-ubuntu-xenial/1f87c5b/ : FAILURE in 15m 56s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/4/check/gate-grenade-dsvm-ubuntu-trusty/909161d/ : FAILURE in 19m 38s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-nova-tox-db-functional-ubuntu-xenial/39988ad/ : FAILURE in 4m 27s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/4/check/gate-grenade-dsvm-multinode/0ae5937/ : FAILURE in 37m 13s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/cfae3ee/ : FAILURE in 12m 34s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9faed0d/ : FAILURE in 20m 22s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/d64eeb6/ : FAILURE in 21m 21s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/b65c8a6/ : FAILURE in 13m 15s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/e2e2da8/ : FAILURE in 23m 28s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/c0da577/ : FAILURE in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"791b0686c5919a0135acfff3c088a4a867063a2d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-18 15:11:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/4/check/tempest-dsvm-full-nfv-xenial/a5ff94d : FAILURE in 37m 44s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/4/check/tempest-dsvm-intel-nfv-xenial/0356445 : FAILURE in 43m 16s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ea0c3c9 : FAILURE in 45m 16s","accounts_in_message":[],"_revision_number":4},{"id":"aadf99a0efbdf3b52bf21d46a15c397d20b6d58b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-18 15:45:53.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/4/ : FAILURE in 1h 19m 33s","accounts_in_message":[],"_revision_number":4},{"id":"91c5ad83cdebfa9f0cd77c5fa572601738f571d8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-18 17:29:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/4/check/dsvm-tempest-nova-network/a30aa75 : SUCCESS in 1h 21m 53s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/4/check/dsvm-tempest-neutron-network/b69da80 : SUCCESS in 1h 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa5bfe532691bf6f282de94fcfdb9891a251f32f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-18 21:37:53.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/369180/4/1139/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"34712e0e07bb6e94f8658523724438d4c3224054","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-09-19 06:47:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"88f02cbb738824d6f6b54a60b7eee49225f71469","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-19 07:11:01.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/369180/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"87075b6358692fedfbdc35ba31b476ba43c0e75c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-09-19 07:28:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/5/check-nova/Nova-ML2-Sriov/892340e : SUCCESS in 37m 02s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/5/check-nova/Nova-MACVTAP-ML2-Sriov/2d76189 : SUCCESS in 37m 02s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"599841dd2e1945a148d76281a346daaf91f95359","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 07:36:11.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/5/ : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":5},{"id":"c195a76e9e763fabf73563e0276b9dc34350f41e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-19 07:55:33.000000000","message":"Patch Set 5:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/5/check/check-ibm-tempest-dsvm-full/7301fc4/ : SUCCESS in 1h 07m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"696925d7e77a7ee2961a0c313b18fe25199200a2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-19 08:07:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/5/check/gate-nova-docs-ubuntu-xenial/b086514//doc/build/html/ : SUCCESS in 2m 28s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-nova-pep8-ubuntu-xenial/58a709b/ : SUCCESS in 4m 21s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-nova-python27-db-ubuntu-xenial/f54f93c/ : SUCCESS in 13m 27s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/5/check/gate-nova-python34-db/dfd557f/ : SUCCESS in 10m 58s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/5/check/gate-nova-python35-db-nv/62e644c/ : SUCCESS in 15m 22s (non-voting)\n- gate-grenade-dsvm-neutron-ubuntu-trusty http://logs.openstack.org/80/369180/5/check/gate-grenade-dsvm-neutron-ubuntu-trusty/68fc1ce/ : SUCCESS in 44m 48s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/51923b9/ : SUCCESS in 1h 03m 55s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-full-ubuntu-xenial/f7c38ab/ : SUCCESS in 1h 01m 13s\n- gate-grenade-dsvm-ubuntu-trusty http://logs.openstack.org/80/369180/5/check/gate-grenade-dsvm-ubuntu-trusty/2ecf2d9/ : SUCCESS in 37m 54s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-nova-tox-db-functional-ubuntu-xenial/c5cbc54/ : SUCCESS in 12m 36s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/80/369180/5/check/gate-grenade-dsvm-multinode/afdb9ad/ : SUCCESS in 40m 20s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-cells-ubuntu-xenial/1e1b1af/ : SUCCESS in 31m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/05934a2/ : SUCCESS in 54m 27s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/39dd2b2/ : SUCCESS in 1h 13m 57s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/15daba4/ : SUCCESS in 1h 18m 35s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/3e7b2be/ : SUCCESS in 29m 21s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/5/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/2a169d6/ : SUCCESS in 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ce7632bac407917ed37dfdb024a0f7ae99450e66","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-19 08:24:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/5/check/dsvm-tempest-xen/be5fcf6 : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":5},{"id":"2843ab4d5b4f035249c0f929e5b745cf50a9bece","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-19 08:39:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/5/check/dsvm-tempest-nova-network/bcd8fab : SUCCESS in 1h 21m 22s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/5/check/dsvm-tempest-neutron-network/1b3897e : SUCCESS in 1h 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa09cd393e1f81c5f3c993ba24a06d68c5678188","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-19 08:41:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/5/check/tempest-dsvm-full-nfv-xenial/4f84d77 : SUCCESS in 1h 53m 27s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/5/check/tempest-dsvm-intel-nfv-xenial/856e024 : SUCCESS in 45m 44s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2439b64 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":5},{"id":"25c9655ff7dc79dc2f773b36365f331f7e5c3143","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-09-19 08:55:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/369180/5/1146/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"37837ac930d0fef04d59bdf0d5f4db4bf8b33a12","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-09-19 11:57:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6a677f2356ee0c0671753b0995e21104cb997793","author":{"_account_id":17920,"name":"Jianghua Wang","email":"jianghua.wang@citrix.com","username":"wjhfresh"},"date":"2016-10-09 03:57:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cac89d5795d3e17877bd017f3fb378530eb9d901","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-01 23:14:14.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"c552794bd74c47e456114a1708539f68350dcada","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2016-11-02 08:29:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"09ce54c01397f5973cf75a6135d806720e82206c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 08:59:33.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/369180/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"5818d11bb041cada8a23db3dc3988877d66705b0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-02 09:12:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/80/369180/6/check-nova/Nova-ML2-Sriov/321bcdf : SUCCESS in 34m 50s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/80/369180/6/check-nova/Nova-MACVTAP-ML2-Sriov/b4ac030 : SUCCESS in 38m 15s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"51a42731d292c54d79ac110e1c780024d3d53266","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 09:20:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/369180/6/check/check-dsvm-tempest-vz7-exe-minimal/dedc8f6 : SUCCESS in 50m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"50b59f1ea9f31bf2cc989d07a6516da0aef3c096","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 09:26:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/80/369180/6/check/dsvm-tempest-kvm/80fe905 : SUCCESS in 56m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"f2d2a8ef48f47f4bdd9e91d8c1a086e1a9be8f0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 09:43:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/6/check/gate-nova-docs-ubuntu-xenial/6070354//doc/build/html/ : SUCCESS in 2m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-nova-pep8-ubuntu-xenial/e24f776/ : SUCCESS in 7m 18s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-nova-python27-db-ubuntu-xenial/b9998b1/ : SUCCESS in 8m 26s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/6/check/gate-nova-python34-db/819cc5c/ : SUCCESS in 7m 07s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/6/check/gate-nova-python35-db-nv/3bfdf30/ : SUCCESS in 7m 27s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6bf7776/ : SUCCESS in 1h 12m 48s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e04ec32/ : SUCCESS in 49m 20s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-full-ubuntu-xenial/63c545e/ : SUCCESS in 39m 53s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-grenade-dsvm-ubuntu-xenial/ffed792/ : SUCCESS in 24m 55s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-nova-tox-db-functional-ubuntu-xenial/da83367/ : SUCCESS in 10m 11s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/80/369180/6/check/gate-grenade-dsvm-neutron-multinode/b828155/ : SUCCESS in 32m 47s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/80/369180/6/check/gate-grenade-dsvm-multinode-live-migration-nv/ef7dd5a/ : SUCCESS in 29m 13s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-cells-ubuntu-xenial/b19c59d/ : SUCCESS in 27m 43s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d738e31/ : SUCCESS in 41m 27s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/882ed52/ : SUCCESS in 48m 05s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/c2ede84/ : SUCCESS in 44m 09s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a375b13/ : SUCCESS in 17m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/e835a01/ : SUCCESS in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dff1a0e69415d785285609f6a2d537bc819aa37d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-02 09:56:32.000000000","message":"Patch Set 6:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/6/check/check-ibm-tempest-dsvm-full/929f8ea/ : SUCCESS in 1h 08m 38s (non-voting)\n- check-ibm-tempest-dsvm-neutron-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/6/check/check-ibm-tempest-dsvm-neutron-full/de28a92/ : FAILURE in 1h 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4cce3fe71acb1315a08677b8745ae3c090da2a6c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-02 09:58:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/6/check/dsvm-tempest-xen/ba59205 : SUCCESS in 1h 27m 53s","accounts_in_message":[],"_revision_number":6},{"id":"1388124056fb81bb47df1c4530f1fe700ad5f6f0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 10:12:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/6/check/tempest-dsvm-full-nfv-xenial/de76a58 : SUCCESS in 1h 42m 13s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/6/check/tempest-dsvm-intel-nfv-xenial/bab861d : SUCCESS in 39m 39s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/da89fe8 : SUCCESS in 54m 21s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/fac62c4 : SUCCESS in 1h 12m 46s (non-voting)\n- tempest-dsbaremetal-pcipassthrough-xenial http://intel-openstack-ci-logs.ovh/80/369180/6/check/tempest-dsbaremetal-pcipassthrough-xenial/b0877b8 : SUCCESS in 41m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c7813179b3e1437df87a08b470c14f179415b89","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-02 10:47:42.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-369180-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"0bdc452e1db60128fa5ca434247d3cead3fd0049","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-02 10:48:31.000000000","message":"Patch Set 6: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/369180/6/1429/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/369180/6/1429/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":6},{"id":"a5b2c09ac48f74fe38c3b4bb5fe780f379bd568b","author":{"_account_id":21239,"name":"Alex Szarka","email":"szarka@inf.u-szeged.hu","username":"xavvior"},"date":"2016-11-02 10:49:52.000000000","message":"Patch Set 6: Code-Review+1\n\nGood patch.\nLooks good","accounts_in_message":[],"_revision_number":6},{"id":"79cfb603ac9a047825478a93514e25b34314772d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-02 11:41:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/6/check/dsvm-tempest-nova-network/73a3ea4 : SUCCESS in 1h 20m 12s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/6/check/dsvm-tempest-neutron-network/a8ae5c1 : SUCCESS in 1h 47m 43s","accounts_in_message":[],"_revision_number":6},{"id":"b654c244e7d104ca47e4e4f7bbb099c2a4bb642b","author":{"_account_id":13404,"name":"Alberto Planas","email":"aplanas@gmail.com","username":"aplanas"},"date":"2016-11-02 12:01:20.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4aff06eab95794b86ea88870d436b506b168328d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2016-11-02 18:53:47.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nSure, looks fine to me, thanks Moshe!\n\n-jay","accounts_in_message":[],"_revision_number":6},{"id":"c632c06df9c377ba3af1b23e2a7365fa66aacce3","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 19:27:25.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/6/ : FAILURE in 39m 23s","accounts_in_message":[],"_revision_number":6},{"id":"a8e78fec3746cd0b90ed5cdb4f11395c2b5a4b82","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2016-11-03 14:20:35.000000000","message":"Patch Set 7: Published edit on patch set 6","accounts_in_message":[],"_revision_number":7},{"id":"d060552820a35e8a2d6787ddeb01965332d7621d","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2016-11-03 14:21:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nApproving after making the docs comment change Jay suggested with his +2","accounts_in_message":[],"_revision_number":7},{"id":"046f03a80c0a11353bbf3208c70db198ef382968","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-03 14:39:47.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/369180/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"61d729e930f5fd3f77c7f99f2f3eed2938b873a0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 15:11:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/80/369180/7/check/check-dsvm-tempest-vz7-exe-minimal/35e0625 : SUCCESS in 50m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"e6033573b7be1042e9dc2b539165921e3d16d4bb","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 15:24:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/80/369180/7/check/dsvm-tempest-kvm/77f8674 : SUCCESS in 1h 03m 36s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"84aded2ff2201d2f2ffb7fe7ce3425a44bfe882b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 15:32:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/7/check/gate-nova-docs-ubuntu-xenial/cb5a329//doc/build/html/ : SUCCESS in 4m 48s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-nova-pep8-ubuntu-xenial/835e0bd/ : SUCCESS in 5m 29s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-nova-python27-db-ubuntu-xenial/64cb74e/ : SUCCESS in 11m 48s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/7/check/gate-nova-python34-db/a3c7587/ : SUCCESS in 8m 56s\n- gate-nova-python35-db-nv http://logs.openstack.org/80/369180/7/check/gate-nova-python35-db-nv/2fe7f44/ : SUCCESS in 9m 38s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/25374f7/ : SUCCESS in 1h 08m 02s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-grenade-dsvm-neutron-ubuntu-xenial/84fbb20/ : SUCCESS in 57m 54s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-grenade-dsvm-ubuntu-xenial/f39c73a/ : SUCCESS in 47m 15s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-nova-tox-db-functional-ubuntu-xenial/03cbc66/ : SUCCESS in 22m 48s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/80/369180/7/check/gate-grenade-dsvm-neutron-multinode/d0bda43/ : SUCCESS in 39m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/80/369180/7/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/1a1532e/ : SUCCESS in 36m 56s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-cells-ubuntu-xenial/a79ce0d/ : SUCCESS in 40m 54s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4dbe952/ : SUCCESS in 48m 19s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/5f28d69/ : SUCCESS in 1h 00m 07s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/9eeb79d/ : SUCCESS in 46m 26s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/95c6907/ : SUCCESS in 23m 36s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/80/369180/7/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/5b78633/ : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a72f9448ddd13015c5e7c1369f8c4b1b6adea800","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 15:32:21.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"581dfd013bcf5b41ba0d611dc48184cd973fcb3a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-03 15:38:28.000000000","message":"Patch Set 7: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/369180/7/1526/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/369180/7/1526/ : FAILURE (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":7},{"id":"fa6a6e7eb68aeccf59dfea7e6b05fe9190f9540c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-03 15:45:09.000000000","message":"Patch Set 7:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/7/check/check-ibm-tempest-dsvm-full/254abcf/ : SUCCESS in 54m 29s (non-voting)\n- check-ibm-tempest-dsvm-neutron-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/80/369180/7/check/check-ibm-tempest-dsvm-neutron-full/f1a6a68/ : SUCCESS in 1h 23m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ecc2f039d4d023f17b67612dfd8d2b83fc532e12","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-03 15:48:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/80/369180/7/check/dsvm-tempest-xen/046fb12 : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":7},{"id":"3d313a7f1b57c961efbfd6183f8df806e47b1ccf","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 16:22:06.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/369180/7/ : SUCCESS in 2h 01m 04s","accounts_in_message":[],"_revision_number":7},{"id":"3655bf6c516d3b4d7173b442ad8b16893f13e128","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-03 17:01:06.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-80-369180-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"701883eccc575cdde4fcefbdf5c67402959f215a","date":"2016-11-03 17:19:54.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":7},{"id":"48db809797db05a8cadaf90a854c36dfaa28beaa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 17:19:54.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/80/369180/7/gate/gate-nova-docs-ubuntu-xenial/daafbb6//doc/build/html/ : SUCCESS in 2m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-nova-pep8-ubuntu-xenial/3a05417/ : SUCCESS in 4m 53s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-nova-python27-db-ubuntu-xenial/16cf41f/ : SUCCESS in 9m 48s\n- gate-nova-python34-db http://logs.openstack.org/80/369180/7/gate/gate-nova-python34-db/3bbf220/ : SUCCESS in 8m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ae0f2bf/ : SUCCESS in 44m 41s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/640ea29/ : SUCCESS in 31m 28s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-grenade-dsvm-ubuntu-xenial/5c0a05c/ : SUCCESS in 26m 11s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-nova-tox-db-functional-ubuntu-xenial/fcf1a92/ : SUCCESS in 12m 15s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/80/369180/7/gate/gate-grenade-dsvm-neutron-multinode/8d1293c/ : SUCCESS in 38m 16s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-tempest-dsvm-cells-ubuntu-xenial/95129c9/ : SUCCESS in 29m 12s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7c22514/ : SUCCESS in 44m 32s\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/80/369180/7/gate/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5d57b19/ : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":7},{"id":"adfa9856aae5bc3f794978de766ef82e9160dd76","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-11-03 18:42:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/7/check/dsvm-tempest-nova-network/95c4516 : SUCCESS in 1h 36m 44s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/80/369180/7/check/dsvm-tempest-neutron-network/bd4d8a6 : SUCCESS in 1h 38m 26s","accounts_in_message":[],"_revision_number":7},{"id":"08ba2290e56ed1d37f7ac6d00613377d8975b525","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-04 03:10:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/7/check/tempest-dsvm-full-nfv-xenial/acc3c53 : SUCCESS in 1h 35m 25s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/80/369180/7/check/tempest-dsvm-intel-nfv-xenial/49e4dc5 : SUCCESS in 35m 37s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/8895320 : SUCCESS in 59m 26s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/80/369180/7/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/cc8f3a5 : SUCCESS in 1h 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6689c96cdfaa148037dc5df2baa058b09a075ec7","revisions":{"5bf7352279fb96abc80c8ce253abf8d55ed9726a":{"kind":"REWORK","_number":1,"created":"2016-09-13 07:22:28.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/1"}}},"commit":{"parents":[{"commit":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","subject":"Fix policy tests for project_id enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110c12b4145a5a978e3dd9e99e65be1d4a86ee06"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:31:06.000000000","tz":180},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bf7352279fb96abc80c8ce253abf8d55ed9726a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bf7352279fb96abc80c8ce253abf8d55ed9726a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","is_merged_in_target_branch":true,"change_id":"I377ef4ef438d5ef7e2c591a8317a02bc99575783","change_number":354690,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b46905f7c9e120baff3d36d4ed5c0a9517dcebb1":{"kind":"REWORK","_number":2,"created":"2016-09-15 11:38:55.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/2"}}},"commit":{"parents":[{"commit":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","subject":"Fix policy tests for project_id enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110c12b4145a5a978e3dd9e99e65be1d4a86ee06"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-15 11:36:48.000000000","tz":180},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b46905f7c9e120baff3d36d4ed5c0a9517dcebb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b46905f7c9e120baff3d36d4ed5c0a9517dcebb1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","is_merged_in_target_branch":true,"change_id":"I377ef4ef438d5ef7e2c591a8317a02bc99575783","change_number":354690,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0189c43c78d48e85b805078474d63ac53fbe59b8":{"kind":"REWORK","_number":3,"created":"2016-09-18 05:50:18.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/3"}}},"commit":{"parents":[{"commit":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","subject":"Fix policy tests for project_id enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110c12b4145a5a978e3dd9e99e65be1d4a86ee06"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-18 05:43:48.000000000","tz":180},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0189c43c78d48e85b805078474d63ac53fbe59b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0189c43c78d48e85b805078474d63ac53fbe59b8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","is_merged_in_target_branch":true,"change_id":"I377ef4ef438d5ef7e2c591a8317a02bc99575783","change_number":354690,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dc791dd7dbc14855c2df1e25d4f1501f12b55f15":{"kind":"REWORK","_number":4,"created":"2016-09-18 14:25:55.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/4"}}},"commit":{"parents":[{"commit":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","subject":"Fix policy tests for project_id enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110c12b4145a5a978e3dd9e99e65be1d4a86ee06"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-18 14:23:10.000000000","tz":180},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc791dd7dbc14855c2df1e25d4f1501f12b55f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc791dd7dbc14855c2df1e25d4f1501f12b55f15"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","is_merged_in_target_branch":true,"change_id":"I377ef4ef438d5ef7e2c591a8317a02bc99575783","change_number":354690,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a57e87facd994f89b697d489199c045edc6f9f16":{"kind":"REWORK","_number":5,"created":"2016-09-19 06:47:19.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/5"}}},"commit":{"parents":[{"commit":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","subject":"Fix policy tests for project_id enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110c12b4145a5a978e3dd9e99e65be1d4a86ee06"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-19 06:44:09.000000000","tz":180},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a57e87facd994f89b697d489199c045edc6f9f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a57e87facd994f89b697d489199c045edc6f9f16"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"110c12b4145a5a978e3dd9e99e65be1d4a86ee06","is_merged_in_target_branch":true,"change_id":"I377ef4ef438d5ef7e2c591a8317a02bc99575783","change_number":354690,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7a68ddd8b96fe756c0ada2617845004746a205a3":{"kind":"REWORK","_number":6,"created":"2016-11-02 08:29:45.000000000","uploader":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"ref":"refs/changes/80/369180/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/6"}}},"commit":{"parents":[{"commit":"40f9b0ad16d3a5fae184bbe6d4a49cf792967089","subject":"Merge \"pci: in free_device(), compare by device id and not reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40f9b0ad16d3a5fae184bbe6d4a49cf792967089"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-11-02 08:28:18.000000000","tz":120},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a68ddd8b96fe756c0ada2617845004746a205a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a68ddd8b96fe756c0ada2617845004746a205a3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40f9b0ad16d3a5fae184bbe6d4a49cf792967089","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6689c96cdfaa148037dc5df2baa058b09a075ec7":{"kind":"REWORK","_number":7,"created":"2016-11-03 14:20:35.000000000","uploader":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"ref":"refs/changes/80/369180/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/369180/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/369180/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/369180/7"}}},"commit":{"parents":[{"commit":"40f9b0ad16d3a5fae184bbe6d4a49cf792967089","subject":"Merge \"pci: in free_device(), compare by device id and not reference\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40f9b0ad16d3a5fae184bbe6d4a49cf792967089"}]}],"author":{"name":"Moshe Levi","email":"moshele@mellanox.com","date":"2016-09-13 06:30:59.000000000","tz":180},"committer":{"name":"Daniel Berrange","email":"berrange@redhat.com","date":"2016-11-03 14:20:27.000000000","tz":0},"subject":"pci: remove pci device from claims and allocations when freeing it","message":"pci: remove pci device from claims and allocations when freeing it\n\nIn drop_move_claim we call free pci device when we need to drop a\nspecific device from the src/dest node. This is done by calling\npci manager free_device. The current code just update the device\nstatus to available in database but doesn\u0027t remove it from the\npci manager claims and allocations lists. This patch adds the\nremoval as well.\n\nCloses-Bug: #1622854\n\nChange-Id: If1cd6f3a635759cd55d116a34ca164630c61e085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6689c96cdfaa148037dc5df2baa058b09a075ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6689c96cdfaa148037dc5df2baa058b09a075ec7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"40f9b0ad16d3a5fae184bbe6d4a49cf792967089","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
