)]}'
{"id":"openstack%2Fnova~378009","triplet_id":"openstack%2Fnova~master~I8cfb2280d956d452ccad1fc711bd814b7258147f","project":"openstack/nova","branch":"master","topic":"bug/1627694","hashtags":[],"change_id":"I8cfb2280d956d452ccad1fc711bd814b7258147f","subject":"Handle spawning error on unshelving","status":"MERGED","created":"2016-09-27 20:06:28.000000000","updated":"2018-02-27 22:09:06.000000000","submitted":"2018-02-27 17:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"378009-1519753948826-b3948aa5","meta_rev_id":"0f346320902e21308a745e75b38ba8f94c3686fb","_number":378009,"virtual_id_number":378009,"owner":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-02-22 03:30:51.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-22 03:37:29.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},{"value":0,"date":"2018-02-22 09:26:42.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-02-22 03:42:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"date":"2018-02-22 06:30:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-02-27 17:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-22 04:48:13.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2018-02-22 09:12:40.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},{"value":0,"date":"2018-02-22 04:34:49.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-23 11:51:05.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":2,"date":"2018-02-22 14:16:58.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":1,"date":"2018-02-26 11:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-02-27 16:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-02-27 16:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-27 20:26:26.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-09-27 21:28:03.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2016-09-28 06:57:39.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-10-11 12:51:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-10-11 13:22:10.000000000","updated_by":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"reviewer":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"state":"REVIEWER"},{"updated":"2016-10-14 15:55:21.000000000","updated_by":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"reviewer":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"state":"REVIEWER"},{"updated":"2017-04-13 19:47:32.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-04-13 21:09:04.000000000","updated_by":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-14 02:58:51.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-07 09:29:45.000000000","updated_by":{"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},"reviewer":{"_account_id":18094,"name":"Shoham Peller","username":"speller","inactive":true},"state":"REVIEWER"},{"updated":"2017-10-07 11:06:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-11-25 08:14:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-01 00:03:26.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-01 03:31:02.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-12-01 14:10:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-18 18:14:36.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2018-02-21 14:22:09.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2018-02-21 23:00:57.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-02-22 02:25:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 03:30:51.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 03:37:29.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 03:42:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 04:34:49.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 04:48:13.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 06:30:32.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-02-22 09:12:40.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-02-22 09:26:42.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-22 14:16:58.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-02-23 11:51:05.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-26 11:58:38.000000000","updated_by":{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},"reviewer":{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},"state":"REVIEWER"},{"updated":"2018-02-27 16:24:51.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-02-27 17:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"106720064e9b4766a84996f3fc30073c03331e80","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2016-09-27 20:06:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d106d2efba12b347c308f35be90654936751ccb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-27 20:07:00.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"347be3dec840601e0f20c8907adcb6008b18a547","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-09-27 20:26:26.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/378009/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a4819379014825ed212b5436f1b7468c7979c014","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-27 21:19:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/09/378009/1/check/gate-nova-docs-ubuntu-xenial/e0943b1//doc/build/html/ : SUCCESS in 2m 31s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-pep8-ubuntu-xenial/55a26a7/ : SUCCESS in 4m 16s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-python27-db-ubuntu-xenial/2907bb0/ : SUCCESS in 15m 35s\n- gate-nova-python34-db http://logs.openstack.org/09/378009/1/check/gate-nova-python34-db/4e78030/ : SUCCESS in 8m 15s\n- gate-nova-python35-db-nv http://logs.openstack.org/09/378009/1/check/gate-nova-python35-db-nv/4d9599c/ : SUCCESS in 8m 54s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/68488bf/ : SUCCESS in 42m 18s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/086d475/ : SUCCESS in 30m 59s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-ubuntu-xenial/32e16ca/ : SUCCESS in 37m 06s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-ubuntu-xenial/daec110/ : SUCCESS in 23m 16s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-tox-db-functional-ubuntu-xenial/909ee15/ : SUCCESS in 10m 32s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-multinode/a66c18d/ : SUCCESS in 43m 37s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/2c3b816/ : SUCCESS in 27m 30s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a4951ef/ : FAILURE in 42m 27s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/faaa617/ : SUCCESS in 1h 12m 39s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/f3b6039/ : SUCCESS in 40m 42s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/78c2d01/ : SUCCESS in 16m 44s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/c0fa868/ : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6ff8d13cd969e6df81db83a5337e71099a5e67d","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-27 21:28:03.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1627220\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"c02bcb8cab6bd2a24e4e0c8f71cc3e1f44165de0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-09-27 21:48:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/09/378009/1/check/dsvm-tempest-xen/27f70b4 : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0e62c9bbf3edc8414b0d79de32abfcf02c1c89cb","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-27 22:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/378009/1/ : SUCCESS in 2h 04m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0c590ae70e940b5bc128b90cac88c8989b3b8e2a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2016-09-27 23:39:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-nova-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/1/check/dsvm-tempest-nova-network/723a82b : SUCCESS in 1h 16m 59s\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/1/check/dsvm-tempest-neutron-network/50ca0a0 : SUCCESS in 1h 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"108f9e6b1d59663280abba54ec82b250ce19afba","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2016-09-28 02:57:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"672d447c8ffb4951a34eeca3b8206df3cff01d6f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-28 02:57:48.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab00963731dab35cfac320e4da3cd2b9049eba42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-28 04:31:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/09/378009/1/check/gate-nova-docs-ubuntu-xenial/16da7e6//doc/build/html/ : SUCCESS in 5m 15s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-pep8-ubuntu-xenial/69a2b90/ : SUCCESS in 6m 13s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-python27-db-ubuntu-xenial/0f0e5ef/ : SUCCESS in 9m 34s\n- gate-nova-python34-db http://logs.openstack.org/09/378009/1/check/gate-nova-python34-db/413360b/ : SUCCESS in 13m 39s\n- gate-nova-python35-db-nv http://logs.openstack.org/09/378009/1/check/gate-nova-python35-db-nv/8838291/ : SUCCESS in 9m 24s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/943960a/ : SUCCESS in 42m 46s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/87bed17/ : SUCCESS in 28m 05s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-ubuntu-xenial/79e3b50/ : SUCCESS in 40m 34s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-ubuntu-xenial/dc704bc/ : SUCCESS in 32m 05s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-tox-db-functional-ubuntu-xenial/5e8edda/ : SUCCESS in 11m 01s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-multinode/6883bd2/ : SUCCESS in 26m 40s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/69f43e4/ : SUCCESS in 27m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/6a193f3/ : FAILURE in 44m 18s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/855961b/ : SUCCESS in 1h 24m 03s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/3e6fe90/ : SUCCESS in 1h 07m 14s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/9793655/ : SUCCESS in 16m 58s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/fd26a66/ : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22c9ad043c79848d68746d545da9095c4b4378dc","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2016-09-28 04:31:34.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1627220\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"767b465dae2773e479187c375a8060aa269db20b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-28 05:05:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-full-nfv-xenial/5b31042 : SUCCESS in 1h 41m 37s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-intel-nfv-xenial/be99c90 : SUCCESS in 34m 22s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4af7837 : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5c8ca1ccc858d72b596f59faaa64518be9e7f3d9","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2016-09-28 05:11:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"32efa3e4ff5221991f6b1c9b0a34eff906f91f93","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-09-28 05:11:53.000000000","message":"Patch Set 1:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full check-ibm-tempest-dsvm-full : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a872ffe45af281a973656ab52db27ddea68049f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-28 05:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/09/378009/1/check/gate-nova-docs-ubuntu-xenial/119d3f1//doc/build/html/ : SUCCESS in 2m 25s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-pep8-ubuntu-xenial/ff2d9f1/ : SUCCESS in 4m 25s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-python27-db-ubuntu-xenial/8fab6b2/ : SUCCESS in 9m 30s\n- gate-nova-python34-db http://logs.openstack.org/09/378009/1/check/gate-nova-python34-db/aac34e6/ : SUCCESS in 11m 49s\n- gate-nova-python35-db-nv http://logs.openstack.org/09/378009/1/check/gate-nova-python35-db-nv/060cdfa/ : SUCCESS in 8m 32s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a3e41b3/ : SUCCESS in 41m 30s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e66485b/ : SUCCESS in 29m 01s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-ubuntu-xenial/2ef9e25/ : SUCCESS in 38m 07s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-ubuntu-xenial/64d62d6/ : SUCCESS in 36m 49s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-tox-db-functional-ubuntu-xenial/279217b/ : SUCCESS in 15m 00s\n- gate-grenade-dsvm-multinode http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-multinode/2b67628/ : SUCCESS in 25m 30s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/2962fa8/ : SUCCESS in 42m 20s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/95bb7f2/ : SUCCESS in 43m 03s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/94b76ef/ : FAILURE in 17m 10s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/e11f423/ : SUCCESS in 41m 06s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/fea98f4/ : SUCCESS in 17m 07s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/f74f155/ : SUCCESS in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da976e0cf97da47b8cf148b3c951fef5421ae090","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-09-28 06:27:01.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-09-378009-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"e0d62e6600b2e2322bac99779c56a2420fc44003","author":{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},"date":"2016-09-28 06:44:08.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"5a9339a13ad6572844fa5ff10455823e6266e049","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-09-28 06:57:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/378009/1/ : FAILURE in 4h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e0a990e98fad5a4d72583acbcbca2ce5d8c19d26","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-09-28 12:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-full-nfv-xenial/d64f703 : SUCCESS in 1h 59m 26s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-intel-nfv-xenial/fcf13ed : SUCCESS in 43m 35s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/09/378009/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/97f1038 : SUCCESS in 1h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8094551ec707aad2f2137eedb15cae6117306eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-01 15:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild Successful \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/378009/1/1686/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b5f40b995d8d2d4bbd3996b0974771db2277316b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-10-01 15:21:55.000000000","message":"Patch Set 1: -Verified\n\nBuild Failed \n\n* vmware-dsvm-ext-nova http://208.91.1.172/logs/vmware-dsvm-ext-nova/378009/1/1687/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"46e55ad9bba213c17c5e4ac90060435ab0be15f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-10-11 12:51:02.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks sane to me","accounts_in_message":[],"_revision_number":1},{"id":"d1a53511576c620b67cdeed6acd839c25189e715","author":{"_account_id":14819,"name":"Timofey Durakov","email":"timofei.nd@gmail.com","username":"tdurakov"},"date":"2016-10-11 13:22:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf079c58e9c3ccd3a3f5e6d098a45f096018e092","author":{"_account_id":19590,"name":"Sarafraj Singh","email":"Sarafraj.Singh@intel.com","username":"sarafrajsingh"},"date":"2016-10-14 15:55:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e711083a298ce39b918e197a2aec4b814303854","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-01-25 05:51:08.000000000","message":"Abandoned\n\nThis patch has been sitting unchanged for more than 12 weeks. I am therefore going to abandon it to keep the review queue sane. Please feel free to restore the change if you\u0027re still working on it.","accounts_in_message":[],"_revision_number":1},{"id":"488cd957e18003cd6d72c33e60c91482ce18c38e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-04-13 18:18:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"562ee4d17f6a77511828711ec51a9376f0d2875f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-04-13 18:19:28.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is a valid bug, but we need a unit test for the fix please.","accounts_in_message":[],"_revision_number":1},{"id":"d8cdb1bf6c144ef887179d21459722545a6a9013","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-04-13 19:08:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/1/check/dsvm-tempest-neutron-network/6b1ae9f : FAILURE in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"76915ec94aecc910f2bbde2c32c08c23f87d4245","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-04-13 19:12:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/1/check/tempest-dsvm-full-xenial/9b2e05f/ : SUCCESS in 52m 45s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/1/check/grenade-dsvm-xenial/99d9182/ : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f22ffb87e06798bd64199a39d04d59b9d21ed05f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-04-13 19:31:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/09/378009/1/check/check-dsvm-tempest-vz7-exe-minimal/93fe19c : SUCCESS in 1h 12m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"ae5cf5b807c337bddf92929cff01416a5e3fc029","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-04-13 19:42:21.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/378009/1/check-ibm-tempest-dsvm-full/c2fb716 : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2108eeed1ebc0dae87d9b8f5c18f9fd8b9f0f614","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-13 19:44:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/09/378009/1/check/gate-nova-docs-ubuntu-xenial/3b1eaa1//doc/build/html/ : SUCCESS in 3m 57s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-pep8-ubuntu-xenial/ad68d5c/ : SUCCESS in 8m 56s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-python27-ubuntu-xenial/cf55035/ : SUCCESS in 10m 40s\n- gate-nova-python35 http://logs.openstack.org/09/378009/1/check/gate-nova-python35/cc3a1e9/ : SUCCESS in 11m 17s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/433f680/ : SUCCESS in 1h 24m 30s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/410751c/ : SUCCESS in 50m 53s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/bc8e2c8/ : SUCCESS in 1h 11m 33s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-tox-functional-ubuntu-xenial/9c7306a/ : SUCCESS in 14m 52s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/8a055a4/ : SUCCESS in 15m 13s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b4894de/ : SUCCESS in 55m 31s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/09/378009/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/e3363ba/ : SUCCESS in 35m 33s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/a2a25c5/ : SUCCESS in 37m 18s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/848db57/ : FAILURE in 28m 55s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e9e2c78/ : SUCCESS in 1h 11m 33s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/7196eff/ : SUCCESS in 1h 05m 44s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/e256a59/ : SUCCESS in 24m 08s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/09/378009/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/83a2a0e/ : SUCCESS in 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcc3aab7b4de4c0e2d63a12b19977d6f47a938a6","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-04-13 19:47:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/378009/1/ : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ee0f8d49ef2a116f08a6db8460a35b1f96783d8f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-04-13 21:00:07.000000000","message":"Patch Set 1: -Verified\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/09/378009/1/check/dsvm-tempest-xen/7fdcb39 : SUCCESS in 2h 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b74c06789c4ec5e63606599fc94b75f493e6135d","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2017-04-13 21:09:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-13/378009/1/check/tempest-dsvm-full-nfv-xenial/f353790 : SUCCESS in 2h 02m 04s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-13/378009/1/check/tempest-dsvm-intel-nfv-xenial/9e71672 : SUCCESS in 50m 57s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-13/378009/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/a02e7a5 : SUCCESS in 1h 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7c1b2d36d63dd74ab04a31f6a254ebbe586b06e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-04-13 22:10:36.000000000","message":"Patch Set 1: -Verified\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-13/378009/1/check/tempest-dsvm-full-nfv-xenial/df3d586 : SUCCESS in 1h 44m 15s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-13/378009/1/check/tempest-dsvm-intel-nfv-xenial/d093009 : SUCCESS in 42m 14s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-04-13/378009/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/14ba971 : SUCCESS in 1h 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89865d801d50fb356e8739787d6c3d2260e0dae9","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-04-14 02:58:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/09/378009/1/check/dsvm-tempest-kvm/062704a : SUCCESS in 1h 20m 20s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/09/378009/1/check/dsvm-tempest-exe/712e5e5 : SUCCESS in 58m 18s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"0d3cacff57a77867508a154cdac19589d3a432b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 18:55:14.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f2b74cba8d562fab27007bf10a2d1b3ce531c14d","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2017-10-07 09:29:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"824c176bdbe7c75bb2a958fa0a3377fecb046ed7","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2017-10-07 09:33:39.000000000","message":"Patch Set 2: Workflow-1 Code-Review-1\n\nWorking on that unit-test","accounts_in_message":[],"_revision_number":2},{"id":"ea49499f9ee1456ed58e4a5f53172d947c7e97e2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-07 10:26:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/2/check/tempest-dsvm-full-xenial/37d3683/ : SUCCESS in 55m 55s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/2/check/grenade-dsvm-xenial/a066ca5/ : SUCCESS in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4915669d1865bc2c0ab12e92df4e00c0bbd1281","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 10:26:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"2a450bba9ad32d79fdb2352d3d7f07e316574476","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:06:04.000000000","message":"Patch Set 2: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/378009/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/b4170b7 : SUCCESS in 1h 35m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b64f8c1249347dcb7f2ac62c3a9ba9b41c0a0a93","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:06:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"a7703f05677c7b3250687886be92f9a4104ef904","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-07 11:06:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/09/378009/2/check/gate-nova-docs-ubuntu-xenial/1bade62//doc/build/html/ : SUCCESS in 5m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-nova-pep8-ubuntu-xenial/7152d64/ : SUCCESS in 8m 05s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-nova-python27-ubuntu-xenial/4d987fd/ : SUCCESS in 12m 07s\n- gate-nova-python35 http://logs.openstack.org/09/378009/2/check/gate-nova-python35/36cdd68/ : SUCCESS in 13m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/49f54b6/ : SUCCESS in 1h 24m 25s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ae7daa4/ : SUCCESS in 51m 10s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/4a2f52b/ : SUCCESS in 1h 16m 04s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-nova-tox-functional-ubuntu-xenial/de94488/ : SUCCESS in 21m 56s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/003ddbc/ : SUCCESS in 17m 33s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/3bd5c3a/ : SUCCESS in 1h 16m 03s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/09/378009/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/742ef55/ : FAILURE in 59m 22s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/e8435ab/ : SUCCESS in 38m 28s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/54b30e6/ : FAILURE in 16m 24s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/0561eae/ : SUCCESS in 1h 30m 45s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/ea3f442/ : SUCCESS in 1h 21m 50s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/018ec3c/ : SUCCESS in 39m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/09/378009/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/a8aa4d4/ : SUCCESS in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74d0f032d9b858714e4b96643196c7d659bf9552","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:07:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"75c1d94a8df069fffb026d8760c366685abe3e63","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:16:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/2/check/nova-out-of-tree-pvm/8095055 : SUCCESS in 1h 46m 26s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/2/check/nova-in-tree-pvm/7a52476 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":2},{"id":"95293dab00fe58843773445a2ea7effe45daf769","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:18:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-07/378009/2/check/tempest-dsvm-full-nfv-xenial/521475b : FAILURE in 1h 48m 14s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-07/378009/2/check/tempest-dsvm-intel-nfv-xenial/ca31f6d : FAILURE in 36m 54s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-07/378009/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/19c00d4 : FAILURE in 1h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36528c4dfe73f34afb5eb159635da9fe917b5b30","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-07 11:37:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/2/check/dsvm-tempest-neutron-network/0b0fe66 : SUCCESS in 2h 07m 14s","accounts_in_message":[],"_revision_number":2},{"id":"bb943ee338335c24eb84a125f24d7de16116fa6f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-07 11:39:07.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/09/378009/2/check/dsvm-tempest-xen/42923a5 : FAILURE in 2h 08m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5c6cbcea1c544a4597eda13ddada5211ce456dce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-07 13:12:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/2/check/openstack-tox-pep8/3457fa0/ : SUCCESS in 8m 35s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/2/check/openstack-tox-py27/7687fd3/ : SUCCESS in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/2/check/openstack-tox-py35/b2b052e/ : SUCCESS in 12m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/2/check/build-openstack-sphinx-docs/de76d81/html/ : SUCCESS in 4m 58s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-py35/fd1b5b1/ : SUCCESS in 1h 07m 20s\n- legacy-nova-tox-functional http://logs.openstack.org/09/378009/2/check/legacy-nova-tox-functional/d72ea07/ : SUCCESS in 18m 35s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/09/378009/2/check/legacy-nova-tox-functional-py35/e616279/ : SUCCESS in 17m 17s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/09/378009/2/check/legacy-grenade-dsvm-neutron-multinode/e9cbae5/ : SUCCESS in 1h 01m 37s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/a0166d9/ : FAILURE in 50m 05s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-cells/1f996f9/ : SUCCESS in 43m 43s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a7152ed/ : FAILURE in 20m 21s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-neutron-multinode-full/ee7b5e8/ : SUCCESS in 1h 24m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-neutron-nova-next-full/04bb0b7/ : SUCCESS in 1h 20m 24s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-multinode-live-migration/1f18d83/ : SUCCESS in 49m 18s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4d27aae/ : SUCCESS in 48m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/09/378009/2/check/legacy-tempest-dsvm-neutron-full/e1ab42f/ : SUCCESS in 1h 28m 59s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/09/378009/2/check/legacy-grenade-dsvm-neutron/4d4d6be/ : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":2},{"id":"271943d34a4d6d666fbbaac669d3330a2ad421d7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-07 13:12:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/09/378009/2/check-nova/Nova-ML2-Sriov/fc4a659 : SUCCESS in 1h 07m 53s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/09/378009/2/check-nova/Nova-MACVTAP-ML2-Sriov/f97a227 : FAILURE in 52m 46s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"abede7d9632aed3bdb14df8de8b52c97bf427d8b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-07 13:12:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"0e11cb63d7cda93909f62a427adae76b10ec5420","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-07 14:26:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/378009/2/ : FAILURE in 3h 20m 41s","accounts_in_message":[],"_revision_number":2},{"id":"415fc7fed0b4d340bb74c4d5d225a014e31d4385","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-07 15:51:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cee3467d47dcb30998e7d8ffd245f12fdfbae6d0","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2017-10-19 10:40:05.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"83e2df233ab470874eac748858d15b199f1d4c87","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2017-10-19 10:40:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cf2f3d552909a1c9477b79359aca218919084dcf","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2017-10-19 10:42:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea0cab7f08feba1f8d05183ffb6c072b6093ecb4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-19 11:34:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/3/check/tempest-dsvm-full-xenial/5244d7a/ : SUCCESS in 53m 28s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/3/check/grenade-dsvm-xenial/7c10aea/ : FAILURE in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e4aef0165e7ed1f140e658ff2c898ae37abced1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 11:35:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"b4c02872527395a580ef40b1d3cc9ae4428c5473","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 11:59:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/09/378009/3/check/check-dsvm-tempest-vz7-exe-minimal/5537531 : SUCCESS in 1h 17m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"fbb53ae2b3c8eb03fa1770dedb3a3051904cb03e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 12:13:49.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/3/check/nova-out-of-tree-pvm/a281081 : FAILURE in 35m 08s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/3/check/nova-in-tree-pvm/d388715 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":3},{"id":"237d44d35e9cf4916159af3a898cc29af992a4a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 12:22:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/3/check/openstack-tox-pep8/6672ab0/ : SUCCESS in 8m 08s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/3/check/openstack-tox-py27/1fa6a5e/ : SUCCESS in 11m 22s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/3/check/openstack-tox-py35/a5fc23f/ : SUCCESS in 12m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/3/check/build-openstack-sphinx-docs/8cd3721/html/ : SUCCESS in 5m 27s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-py35/73bd0de/ : SUCCESS in 1h 19m 18s\n- legacy-nova-tox-functional http://logs.openstack.org/09/378009/3/check/legacy-nova-tox-functional/3129cc0/ : SUCCESS in 22m 25s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/09/378009/3/check/legacy-nova-tox-functional-py35/72ae881/ : SUCCESS in 20m 18s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/09/378009/3/check/legacy-grenade-dsvm-neutron-multinode/83336c2/ : SUCCESS in 1h 07m 33s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/cad6453/ : FAILURE in 1h 06m 44s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-cells/c282262/ : SUCCESS in 42m 43s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/08905e6/ : SUCCESS in 1h 03m 50s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-neutron-multinode-full/37ddf96/ : SUCCESS in 1h 37m 52s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-neutron-nova-next-full/478834a/ : SUCCESS in 1h 19m 03s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-multinode-live-migration/9173006/ : SUCCESS in 43m 04s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/21446ff/ : SUCCESS in 47m 40s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/09/378009/3/check/legacy-tempest-dsvm-neutron-full/a55d352/ : SUCCESS in 1h 29m 21s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/09/378009/3/check/legacy-grenade-dsvm-neutron/0895c2c/ : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":3},{"id":"38d7995b9c47a6c9a643f685df905ad30f50ea71","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 12:22:26.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"26da3ce3ff218f392bbba1ee67494245eae52e96","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-19 12:32:39.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7398 : FAILURE in 1h 51m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"635aefa6461428e81eff905fb059e5ace87e043b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-19 12:59:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/3/check/dsvm-tempest-neutron-network/84534b9 : SUCCESS in 2h 00m 18s","accounts_in_message":[],"_revision_number":3},{"id":"783e067e753443c70e031660f455371316ebc3ee","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 13:04:26.000000000","message":"Patch Set 3: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/288109a : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a5f5cead3d55ad5c37fc71cebc86afb5a230444f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 13:04:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"d581acebfdd4ba7879d6846657c59d26852b0198","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 13:53:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/378009/3/ : SUCCESS in 2h 03m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ee3ce3d8f47e878a5343809cc5b94653095157d4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-19 18:27:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/09/378009/3/check-nova/Nova-ML2-Sriov/04e8805 : SUCCESS in 57m 17s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/09/378009/3/check-nova/Nova-MACVTAP-ML2-Sriov/bb58ac7 : SUCCESS in 56m 28s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"06f716b620655acd279a74beb5ade2ac2f261b5b","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2017-11-25 06:09:07.000000000","message":"Patch Set 3:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":3},{"id":"e3521d6856a19aa857b02f4c0c34df4f0bda136f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-25 08:13:51.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-9588 : SUCCESS in 1h 58m 36s","accounts_in_message":[],"_revision_number":3},{"id":"412e3302c6f4c4e61c1b6df14f9269137fdcc798","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-11-25 08:14:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/378009/3/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"9011bad0c5765ddd2c7072c508111bc3a9a64f04","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-11-27 03:34:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nsome unit test might need to be added as well","accounts_in_message":[],"_revision_number":3},{"id":"08894aa2ac886426f2d5cb9971b908e3e348100a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 21:14:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2879836f96cfda9eff8edc25e7799d30c28f94dc","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 21:19:27.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7f771d9da3965f31752d68ceef3f58b3e3c1d579","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 21:49:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d335163a8cf244f2634b17cd74b71c03884a7ea8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-11-30 22:46:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/4/check/tempest-dsvm-full-xenial/7b30a4a/ : SUCCESS in 56m 47s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/4/check/grenade-dsvm-xenial/5528469/ : SUCCESS in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fcd419830d420ff5fdff3c01e06e9f1bc4251ab7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-11-30 23:25:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/09/378009/4/check/check-dsvm-tempest-vz7-exe-minimal/644dea6 : SUCCESS in 50m 06s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"2925b8a580fc7f35342ca79364965b9cfd10f591","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 23:34:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/4/check/openstack-tox-pep8/a2efe87/ : SUCCESS in 9m 16s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/4/check/openstack-tox-py27/2c1e324/ : SUCCESS in 12m 13s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/4/check/openstack-tox-py35/6bb311a/ : SUCCESS in 11m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/4/check/build-openstack-sphinx-docs/379e725/html/ : SUCCESS in 5m 16s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-py35/cb8795c/ : SUCCESS in 1h 30m 31s\n- openstack-tox-functional http://logs.openstack.org/09/378009/4/check/openstack-tox-functional/0fa2423/ : SUCCESS in 23m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/09/378009/4/check/openstack-tox-functional-py35/0990d78/ : SUCCESS in 17m 52s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/09/378009/4/check/legacy-grenade-dsvm-neutron-multinode/3848571/ : SUCCESS in 51m 08s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b62d3a0/ : SUCCESS in 58m 30s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-cells/2d9034e/ : SUCCESS in 38m 47s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d66f8e4/ : SUCCESS in 1h 08m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-neutron-multinode-full/f2fa198/ : SUCCESS in 1h 38m 36s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-neutron-nova-next-full/0995ea0/ : SUCCESS in 1h 24m 54s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-multinode-live-migration/01c2010/ : SUCCESS in 32m 42s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f9df784/ : SUCCESS in 39m 16s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/09/378009/4/check/legacy-tempest-dsvm-neutron-full/18c16d0/ : SUCCESS in 1h 21m 12s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/09/378009/4/check/legacy-grenade-dsvm-neutron/613e379/ : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":4},{"id":"e857d7a1d1ff9000b29d3b21f183606adc097366","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-01 00:03:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/09/378009/4/check/dsvm-tempest-xen/ba78b7b : SUCCESS in 2h 13m 27s","accounts_in_message":[],"_revision_number":4},{"id":"3c5b2113959ef62cf62be0ed9367829b9f044a6e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-01 00:16:03.000000000","message":"Patch Set 4: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/6a4989b : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":4},{"id":"d3f010bfb5d67b8b9b16891c3ca528ab577d1de0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-01 00:17:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/4/check/nova-out-of-tree-pvm/cf7fa87 : SUCCESS in 51m 27s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/4/check/nova-in-tree-pvm/d1f957b : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":4},{"id":"893db7712a002cd3c33e8cc73e9c26b7c8c1c5eb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-01 00:50:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://52.169.200.208/09/378009/4/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/7967ccc : SUCCESS in 1h 14m 48s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"1254b17f6c9d5df0625441c05880c5019a139127","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-01 01:52:01.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-09-378009-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"3b6f963a4b5830c990737dfb492f580d47801e87","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-01 02:52:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/09/378009/4/check-nova/Nova-ML2-Sriov/8b7211d : SUCCESS in 1h 28m 21s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/09/378009/4/check-nova/Nova-MACVTAP-ML2-Sriov/efa88e2 : FAILURE in 52m 27s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"e7976b38726a9568e756e451474174e2abd59c76","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-12-01 03:31:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a16c5597b22ef00c4d3c157e30520eec4a6e1de9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-01 08:14:27.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/378009/4/ : FAILURE in 21s","accounts_in_message":[],"_revision_number":4},{"id":"77dcb43d22064217406c81c39907874347f0b3c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-01 14:10:36.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/09/378009/4/check-vote/ext-nova-zuul/f836cac : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"3f9ed4933b08e9ce6f5ca3c491720c87d5b1dd9e","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2017-12-08 21:19:40.000000000","message":"Patch Set 4:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":4},{"id":"ba237d1fb12259a1cddb25acc0e685eeb2b79f8e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-08 22:20:04.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10300 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4d2ac1aa78359a6b9016e6d645d5745756252896","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-12-18 18:14:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5015f5c0b42bf8adf6bd32a31ed31066f975426b","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2018-02-21 11:23:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f2c233efe218b208fb75a049d21208bf24a47664","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2018-02-21 11:31:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6edfa85d0a6400c03afa3e159e17f6546160fa9a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-21 12:23:22.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14464 : FAILURE in 51m 44s","accounts_in_message":[],"_revision_number":6},{"id":"980aa0789bc8bbc8b5cf76e5c1f54004527adb12","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-21 12:27:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/6/check/nova-out-of-tree-pvm/5dc5f06 : SUCCESS in 55m 25s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/6/check/nova-in-tree-pvm/805b1d6 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":6},{"id":"5f902a05c89a5c4d99dfcca1f48c70566449d0b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-21 12:35:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/6/check/tempest-dsvm-full-xenial/c58257b/ : SUCCESS in 1h 03m 30s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/6/check/grenade-dsvm-xenial/9aae7f6/ : SUCCESS in 46m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a152ce2a426a2185892a7895fb796e3f7990e95e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-21 12:55:05.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/95893a3 : SUCCESS in 1h 23m 16s","accounts_in_message":[],"_revision_number":6},{"id":"58629ea7989d90ed1b9ed928a3a8dba3b43e12ac","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-02-21 12:56:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://http://52.27.155.124//portland/2018-02-21/378009/6/check/tempest-dsvm-intel-nfv-xenial/af9f67e : SUCCESS in 42m 01s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-21/378009/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e46c29e : SUCCESS in 1h 23m 27s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-21/378009/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4d74344 : SUCCESS in 1h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5ef71ba85588c22a0d47d272dbc7595f4837673a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 13:20:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/6/check/openstack-tox-pep8/01ed67e/ : SUCCESS in 8m 38s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/6/check/openstack-tox-py27/94f55f8/ : FAILURE in 11m 54s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/6/check/openstack-tox-py35/8435754/ : FAILURE in 12m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/6/check/build-openstack-sphinx-docs/9479531/html/ : SUCCESS in 6m 48s\n- tempest-full http://logs.openstack.org/09/378009/6/check/tempest-full/8b4cc59/ : SUCCESS in 1h 19m 00s\n- neutron-grenade http://logs.openstack.org/09/378009/6/check/neutron-grenade/36228f5/ : SUCCESS in 47m 32s\n- tempest-full-py3 http://logs.openstack.org/09/378009/6/check/tempest-full-py3/05e7476/ : SUCCESS in 1h 20m 37s\n- neutron-grenade-multinode http://logs.openstack.org/09/378009/6/check/neutron-grenade-multinode/d9b6e2a/ : SUCCESS in 56m 32s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/6/check/legacy-grenade-dsvm-neutron-multinode-live-migration/da7ca48/ : SUCCESS in 54m 09s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/6/check/legacy-tempest-dsvm-cells/0d5c30b/ : SUCCESS in 44m 23s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/6/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/528bf42/ : SUCCESS in 59m 32s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/09/378009/6/check/neutron-tempest-multinode-full/c90ee5c/ : SUCCESS in 1h 42m 20s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/6/check/legacy-tempest-dsvm-multinode-live-migration/4486535/ : FAILURE in 36m 08s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/6/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c8be3c4/ : SUCCESS in 35m 38s (non-voting)\n- nova-multiattach http://logs.openstack.org/09/378009/6/check/nova-multiattach/21c6c12/ : SUCCESS in 47m 12s\n- nova-tox-functional http://logs.openstack.org/09/378009/6/check/nova-tox-functional/ff6c93a/ : SUCCESS in 22m 07s\n- nova-tox-functional-py35 http://logs.openstack.org/09/378009/6/check/nova-tox-functional-py35/d59d1af/ : SUCCESS in 23m 44s\n- nova-next http://logs.openstack.org/09/378009/6/check/nova-next/5c2bbe4/ : SUCCESS in 1h 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3a93a5ca4111d4f0c5dc7a81cc379990e904ff77","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-21 13:21:22.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/378009/6/ : FAILURE in 1h 49m 41s","accounts_in_message":[],"_revision_number":6},{"id":"dedf7d5dcb3dbbcd137f64cb950413489ad4f876","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-21 13:41:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/6/check/dsvm-tempest-neutron-network/c89c7ad : SUCCESS in 2h 09m 15s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/6/check/test-vgpu/6684e8b : SUCCESS in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"08f76efcd8661db5c95b852b068c0da789a9042f","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2018-02-21 13:44:03.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ef8d7b89d246ad8b274814dd7d347f7f79dda358","author":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"date":"2018-02-21 13:46:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5e4806491aa64649102b3d5988cbabd4015431a6","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2018-02-21 14:22:09.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fde043bbb7fbd2df707135de53981557cff778f6","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-21 14:40:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/7/check/nova-out-of-tree-pvm/3a53758 : SUCCESS in 53m 10s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/7/check/nova-in-tree-pvm/2c1971f : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":7},{"id":"9966482002b1408997cbe297ec97361a6d8dcab4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-21 14:40:37.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14467 : FAILURE in 53m 18s","accounts_in_message":[],"_revision_number":7},{"id":"ef5c2e5401970b43bb47e936a16b9e01f61d64af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-21 14:48:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/7/check/tempest-dsvm-full-xenial/2b982be/ : SUCCESS in 1h 01m 20s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/7/check/grenade-dsvm-xenial/20d8018/ : SUCCESS in 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0fbfa6c7f39dcf0ae03831be3103e7af6742da68","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-02-21 15:06:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://http://52.27.155.124//portland/2018-02-21/378009/7/check/tempest-dsvm-intel-nfv-xenial/216fb09 : SUCCESS in 47m 58s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-21/378009/7/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/d81b696 : SUCCESS in 1h 15m 40s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://http://52.27.155.124//portland/2018-02-21/378009/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/bf7e1c9 : SUCCESS in 1h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f2420aff007c78cfccce013528ff066de2567a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 15:28:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/7/check/openstack-tox-pep8/79c998e/ : SUCCESS in 8m 08s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/7/check/openstack-tox-py27/2c6df30/ : SUCCESS in 11m 37s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/7/check/openstack-tox-py35/6c5940a/ : SUCCESS in 11m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/7/check/build-openstack-sphinx-docs/777c4b5/html/ : SUCCESS in 5m 32s\n- tempest-full http://logs.openstack.org/09/378009/7/check/tempest-full/47c4ecf/ : SUCCESS in 1h 13m 13s\n- neutron-grenade http://logs.openstack.org/09/378009/7/check/neutron-grenade/6326d77/ : SUCCESS in 50m 53s\n- tempest-full-py3 http://logs.openstack.org/09/378009/7/check/tempest-full-py3/11b2091/ : SUCCESS in 1h 16m 59s\n- neutron-grenade-multinode http://logs.openstack.org/09/378009/7/check/neutron-grenade-multinode/f9eda9b/ : SUCCESS in 1h 03m 29s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/7/check/legacy-grenade-dsvm-neutron-multinode-live-migration/a992ad9/ : SUCCESS in 1h 24m 10s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/7/check/legacy-tempest-dsvm-cells/b61d237/ : SUCCESS in 36m 34s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/7/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/0150416/ : SUCCESS in 1h 05m 31s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/09/378009/7/check/neutron-tempest-multinode-full/4895beb/ : SUCCESS in 1h 33m 36s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/7/check/legacy-tempest-dsvm-multinode-live-migration/3867018/ : SUCCESS in 48m 23s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/7/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0cf3787/ : SUCCESS in 42m 27s (non-voting)\n- nova-multiattach http://logs.openstack.org/09/378009/7/check/nova-multiattach/84c43d4/ : SUCCESS in 43m 04s\n- nova-tox-functional http://logs.openstack.org/09/378009/7/check/nova-tox-functional/9c98e6b/ : SUCCESS in 21m 20s\n- nova-tox-functional-py35 http://logs.openstack.org/09/378009/7/check/nova-tox-functional-py35/8e70ff9/ : SUCCESS in 21m 57s\n- nova-next http://logs.openstack.org/09/378009/7/check/nova-next/3795d75/ : SUCCESS in 1h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"54e37bdfdea1c6e1461b46e21c1f375589b10ad5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-21 15:39:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/09/378009/7/check/check-dsvm-tempest-vz7-exe-minimal/1101b81 : SUCCESS in 1h 13m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"dcaa4069f836ffaac495ce6a15178081010952f2","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-21 15:46:11.000000000","message":"Patch Set 7:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ac5844b : FAILURE in 1h 58m 46s","accounts_in_message":[],"_revision_number":7},{"id":"4b477d500f74550b177a11822cadf53b01893478","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-21 15:49:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/7/check/dsvm-tempest-neutron-network/48631c3 : SUCCESS in 2h 02m 10s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/7/check/test-vgpu/2222c11 : SUCCESS in 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fae72bfd087005f97663e6d2cde713bc68fad357","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-02-21 16:02:15.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a0ac47c2333be96eee9096de9910c28766a4dc99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-02-21 16:13:04.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-09-378009-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"505407cf1d1d4fd1eb7e94d0dd28eaff09edba1a","author":{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},"date":"2018-02-21 16:15:30.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5fc67819abca127d88f85dcf8052fe88f18716d8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-21 16:32:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/378009/7/ : SUCCESS in 2h 45m 33s","accounts_in_message":[],"_revision_number":7},{"id":"2ad302967e8fd84ccb0b678500e5f70b5b00be58","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-02-21 23:00:57.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cc70a9e340e31b5b3ed27e88f3254cb0a52fcba3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-02-22 02:25:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/09/378009/7/check-nova/Nova-ML2-Sriov/001525f : SUCCESS in 1h 03m 55s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/09/378009/7/check-nova/Nova-MACVTAP-ML2-Sriov/778f4cd : SUCCESS in 53m 34s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"226e1908160180554f3b51a99356d8ec6f3d469e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-02-22 02:39:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cc3d7efd2375d874a4353c1c32e0c8b05499efa5","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-22 03:30:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/09/378009/8/check/nova-out-of-tree-pvm/c47ab1b : SUCCESS in 50m 38s\n- nova-in-tree-pvm http://184.172.12.213/09/378009/8/check/nova-in-tree-pvm/e3f5871 : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":8},{"id":"c42829aa74d6458f9727a6098bdd23538416817f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-22 03:33:55.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14501 : FAILURE in 53m 35s","accounts_in_message":[],"_revision_number":8},{"id":"6c623fa54d3d8a353fe9a51cf3f13c7dbf3e02b8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-22 03:37:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/09/378009/8/check/check-dsvm-tempest-vz7-exe-minimal/4425747 : SUCCESS in 56m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"bb04fc95aa372d61c8959fb104040bfe3307a0dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-22 03:42:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/8/check/tempest-dsvm-full-xenial/9394838/ : SUCCESS in 1h 01m 36s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/8/check/grenade-dsvm-xenial/cf1bfa5/ : SUCCESS in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"09b7db69809ed5085c5bc06f1b1cb005b6930eb9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-22 03:58:27.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e458ce3 : FAILURE in 1h 18m 14s","accounts_in_message":[],"_revision_number":8},{"id":"7a069e54e9a9a6961b9d1925000106805591d08d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-22 04:12:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/8/check/openstack-tox-pep8/3d5f297/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/8/check/openstack-tox-py27/25f8451/ : SUCCESS in 10m 39s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/8/check/openstack-tox-py35/d08ecf9/ : SUCCESS in 11m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/8/check/build-openstack-sphinx-docs/0327601/html/ : SUCCESS in 5m 04s\n- tempest-full http://logs.openstack.org/09/378009/8/check/tempest-full/044f8ad/ : SUCCESS in 1h 27m 58s\n- neutron-grenade http://logs.openstack.org/09/378009/8/check/neutron-grenade/f73442c/ : SUCCESS in 44m 04s\n- tempest-full-py3 http://logs.openstack.org/09/378009/8/check/tempest-full-py3/82470ef/ : SUCCESS in 1h 14m 37s\n- neutron-grenade-multinode http://logs.openstack.org/09/378009/8/check/neutron-grenade-multinode/3de189c/ : SUCCESS in 58m 00s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/8/check/legacy-grenade-dsvm-neutron-multinode-live-migration/87b90d3/ : SUCCESS in 1h 01m 50s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-cells/40ec85a/ : SUCCESS in 40m 06s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/e1d22d7/ : SUCCESS in 1h 03m 42s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/09/378009/8/check/neutron-tempest-multinode-full/33b1218/ : SUCCESS in 1h 30m 50s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-multinode-live-migration/132ca24/ : SUCCESS in 40m 42s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/8/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/fc6c2a2/ : SUCCESS in 1h 03m 08s (non-voting)\n- nova-multiattach http://logs.openstack.org/09/378009/8/check/nova-multiattach/c07bec7/ : SUCCESS in 46m 21s\n- nova-tox-functional http://logs.openstack.org/09/378009/8/check/nova-tox-functional/3a732d4/ : SUCCESS in 24m 35s\n- nova-tox-functional-py35 http://logs.openstack.org/09/378009/8/check/nova-tox-functional-py35/2577a65/ : SUCCESS in 23m 50s\n- nova-next http://logs.openstack.org/09/378009/8/check/nova-next/841aa32/ : SUCCESS in 1h 16m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e5d3a2756eea155ce1ee2914946b6be820e4e2b9","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-02-22 04:18:52.000000000","message":"Patch Set 8:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":8},{"id":"95bbe07db45a03e5b4cef5091eac0d853e72d75b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-02-22 04:34:49.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-09-378009-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"7061263bde48d07638d511bd4fbf56333f7fe28d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-22 04:48:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/8/check/dsvm-tempest-neutron-network/80c7dd9 : SUCCESS in 2h 07m 49s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/09/378009/8/check/test-vgpu/459fa26 : SUCCESS in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"873cb09bc2e0070cf31093f4f1ab8ab788caafde","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-22 06:21:19.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/378009/8/ : FAILURE in 3h 41m 07s","accounts_in_message":[],"_revision_number":8},{"id":"4ca85c2139370c6915b3edd46df790a9289462c3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-22 06:30:32.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14505 : SUCCESS in 2h 11m 20s","accounts_in_message":[],"_revision_number":8},{"id":"3520a6d7d48595c72e8afa3546064c10ec5d847f","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2018-02-22 07:41:43.000000000","message":"Patch Set 8:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":8},{"id":"449845c5364652e1ecdd25e3a7d7af4db317a580","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-22 08:56:32.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/8/check/tempest-dsvm-full-xenial/2c0fefc/ : SUCCESS in 1h 08m 48s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/09/378009/8/check/grenade-dsvm-xenial/bafc660/ : SUCCESS in 41m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c58430cabc39cc18abd364e6693151c4b6286ae5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-22 09:12:40.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/378009/8/ : FAILURE in 1h 28m 39s","accounts_in_message":[],"_revision_number":8},{"id":"2ca783aa01959ddda49133b0045de2bc74535fdb","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-22 09:26:42.000000000","message":"Patch Set 8:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/378009/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/829402f : SUCCESS in 1h 39m 44s","accounts_in_message":[],"_revision_number":8},{"id":"25b30a12dab5858b8e3aef295c50567b237acf5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-22 09:41:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/8/check/openstack-tox-pep8/2f64f0d/ : SUCCESS in 7m 38s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/8/check/openstack-tox-py27/7c6e237/ : SUCCESS in 11m 23s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/8/check/openstack-tox-py35/92e46ff/ : SUCCESS in 12m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/8/check/build-openstack-sphinx-docs/33ed189/html/ : SUCCESS in 5m 46s\n- tempest-full http://logs.openstack.org/09/378009/8/check/tempest-full/55a2407/ : SUCCESS in 1h 22m 32s\n- neutron-grenade http://logs.openstack.org/09/378009/8/check/neutron-grenade/a6db5aa/ : SUCCESS in 53m 14s\n- tempest-full-py3 http://logs.openstack.org/09/378009/8/check/tempest-full-py3/1fba090/ : SUCCESS in 1h 35m 59s\n- neutron-grenade-multinode http://logs.openstack.org/09/378009/8/check/neutron-grenade-multinode/c1a34a9/ : SUCCESS in 57m 32s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/09/378009/8/check/legacy-grenade-dsvm-neutron-multinode-live-migration/dd25900/ : SUCCESS in 57m 57s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-cells/e24a33b/ : SUCCESS in 38m 42s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d6ab423/ : SUCCESS in 1h 06m 27s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/09/378009/8/check/neutron-tempest-multinode-full/f2dd0d3/ : SUCCESS in 1h 54m 15s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/8/check/legacy-tempest-dsvm-multinode-live-migration/884a02c/ : SUCCESS in 38m 16s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/09/378009/8/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a111e06/ : SUCCESS in 39m 44s (non-voting)\n- nova-multiattach http://logs.openstack.org/09/378009/8/check/nova-multiattach/21112d4/ : SUCCESS in 1h 03m 39s\n- nova-tox-functional http://logs.openstack.org/09/378009/8/check/nova-tox-functional/d189e89/ : SUCCESS in 19m 39s\n- nova-tox-functional-py35 http://logs.openstack.org/09/378009/8/check/nova-tox-functional-py35/fc37f1f/ : SUCCESS in 22m 42s\n- nova-next http://logs.openstack.org/09/378009/8/check/nova-next/d4c9aa1/ : SUCCESS in 1h 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c25b507eb6a7e99052b6798e5a1e3f6e66ff2e08","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-02-22 14:16:58.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)\n\n++","accounts_in_message":[],"_revision_number":8},{"id":"c752150f64e24bb0fae587c9826c58c164140557","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-02-23 11:51:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-02-23/378009/8/check/tempest-dsvm-intel-nfv-xenial/af54ecd : SUCCESS in 28m 45s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-02-23/378009/8/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/60cf40d : ABORTED in 3m 20s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-02-23/378009/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/12ab1c4 : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c60e95c6b65bc163926860f39f2a115297c93971","author":{"_account_id":20424,"name":"Eyal Posener","email":"eyal@stratoscale.com","username":"eyal"},"date":"2018-02-26 11:58:38.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e775169bb59134dd630c9ea69fdf21a321a4f839","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-02-27 16:24:51.000000000","message":"Patch Set 8: Workflow+1 Code-Review+2\n\n(1 comment)\n\nThis looks like the correct thing to do.","accounts_in_message":[],"_revision_number":8},{"id":"1d04989664983bd5cb46c95a9893e665c2ca4e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-27 16:27:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"798abcba4af77f8486cf0f4b0750288e51dbe56d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-27 17:52:28.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/378009/8/gate/openstack-tox-pep8/3418b7d/ : SUCCESS in 7m 57s\n- openstack-tox-py27 http://logs.openstack.org/09/378009/8/gate/openstack-tox-py27/653cf39/ : SUCCESS in 11m 40s\n- openstack-tox-py35 http://logs.openstack.org/09/378009/8/gate/openstack-tox-py35/c017cf5/ : SUCCESS in 11m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/378009/8/gate/build-openstack-sphinx-docs/4bf82dd/html/ : SUCCESS in 5m 41s\n- tempest-full http://logs.openstack.org/09/378009/8/gate/tempest-full/b3055ad/ : SUCCESS in 1h 21m 12s\n- neutron-grenade http://logs.openstack.org/09/378009/8/gate/neutron-grenade/9381cfa/ : SUCCESS in 50m 49s\n- tempest-full-py3 http://logs.openstack.org/09/378009/8/gate/tempest-full-py3/cc989f8/ : SUCCESS in 1h 19m 34s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/09/378009/8/gate/legacy-tempest-dsvm-cells/c415e07/ : SUCCESS in 39m 55s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/09/378009/8/gate/legacy-tempest-dsvm-multinode-live-migration/41cf754/ : SUCCESS in 35m 16s\n- nova-multiattach http://logs.openstack.org/09/378009/8/gate/nova-multiattach/829f8a9/ : SUCCESS in 43m 40s\n- nova-tox-functional http://logs.openstack.org/09/378009/8/gate/nova-tox-functional/910a3c7/ : SUCCESS in 20m 50s\n- nova-tox-functional-py35 http://logs.openstack.org/09/378009/8/gate/nova-tox-functional-py35/2c080c8/ : SUCCESS in 23m 57s","accounts_in_message":[],"_revision_number":8},{"id":"bba7059ef10f75235790d94fdc65d91ca0825ae9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-27 17:52:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"85a93e61a47f8e5d37b874804f6f75087241f4da","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-02-27 22:09:06.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 73c3e4969c108f2eefc77b57021d9b3e17afdc8e","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"dcdd2c9832c7c60fe9163cd744ca2b5acfe16bcc","revisions":{"448133a078d57d35d06519ae9d9d5fd4fdcdcdae":{"kind":"REWORK","_number":1,"created":"2016-09-27 20:06:28.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/1"}}},"commit":{"parents":[{"commit":"7b2cb0dc0324771d38058c50a8dad62fc3605b6d","subject":"Merge \"Add functional regression test for bug 1595962\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b2cb0dc0324771d38058c50a8dad62fc3605b6d"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 20:06:25.000000000","tz":180},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/448133a078d57d35d06519ae9d9d5fd4fdcdcdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/448133a078d57d35d06519ae9d9d5fd4fdcdcdae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7b2cb0dc0324771d38058c50a8dad62fc3605b6d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"69db9663c65dc4be9f74f78bf9e24d1f84640bda":{"kind":"REWORK","_number":2,"created":"2017-10-07 09:29:45.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/2"}}},"commit":{"parents":[{"commit":"51e16382301ab09f27e2b43db20de86df5709341","subject":"Merge \"Stop joining on system_metadata when listing instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51e16382301ab09f27e2b43db20de86df5709341"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shohamp@gmail.com","date":"2017-10-07 09:05:11.000000000","tz":180},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69db9663c65dc4be9f74f78bf9e24d1f84640bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69db9663c65dc4be9f74f78bf9e24d1f84640bda"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51e16382301ab09f27e2b43db20de86df5709341","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"768d6fd038b26b3aeee752c2b15919f783416a57":{"kind":"REWORK","_number":3,"created":"2017-10-19 10:40:39.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/3"}}},"commit":{"parents":[{"commit":"51e16382301ab09f27e2b43db20de86df5709341","subject":"Merge \"Stop joining on system_metadata when listing instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51e16382301ab09f27e2b43db20de86df5709341"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shohamp@gmail.com","date":"2017-10-19 10:40:11.000000000","tz":180},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/768d6fd038b26b3aeee752c2b15919f783416a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/768d6fd038b26b3aeee752c2b15919f783416a57"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51e16382301ab09f27e2b43db20de86df5709341","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"487a7a78f1116b938b347abb85a6fd28264e3872":{"kind":"REWORK","_number":4,"created":"2017-11-30 21:49:18.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/09/378009/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/4"}}},"commit":{"parents":[{"commit":"100a194ab1ae00cdf0883b1e699a38a626dca549","subject":"Merge \"Remove setting of version/release from releasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/100a194ab1ae00cdf0883b1e699a38a626dca549"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-11-30 21:48:34.000000000","tz":-300},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry.\n\nCo-Authored-By: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/487a7a78f1116b938b347abb85a6fd28264e3872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/487a7a78f1116b938b347abb85a6fd28264e3872"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"100a194ab1ae00cdf0883b1e699a38a626dca549","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1e648f1d97067e21879325ac5b351e208928026b":{"kind":"REWORK","_number":5,"created":"2018-02-21 11:23:33.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/5"}}},"commit":{"parents":[{"commit":"addbfae6769c1033c618019868314ef3100746d9","subject":"Merge \"Detach volumes when deleting a BFV server pre-scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/addbfae6769c1033c618019868314ef3100746d9"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shohamp@gmail.com","date":"2018-02-21 11:22:32.000000000","tz":120},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e648f1d97067e21879325ac5b351e208928026b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e648f1d97067e21879325ac5b351e208928026b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"addbfae6769c1033c618019868314ef3100746d9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"404320171d956c5c9d2b1b07d019151a997f98f8":{"kind":"REWORK","_number":6,"created":"2018-02-21 11:31:13.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/6"}}},"commit":{"parents":[{"commit":"addbfae6769c1033c618019868314ef3100746d9","subject":"Merge \"Detach volumes when deleting a BFV server pre-scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/addbfae6769c1033c618019868314ef3100746d9"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shohamp@gmail.com","date":"2018-02-21 11:30:50.000000000","tz":120},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry.\n\nCo-Authored-By: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/404320171d956c5c9d2b1b07d019151a997f98f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/404320171d956c5c9d2b1b07d019151a997f98f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"addbfae6769c1033c618019868314ef3100746d9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2d162c1d7160628a3ecd78bb5d5470027168bee3":{"kind":"REWORK","_number":7,"created":"2018-02-21 13:46:52.000000000","uploader":{"_account_id":20354,"name":"Shoham Peller","email":"shohamp@gmail.com","username":"shoham.peller"},"ref":"refs/changes/09/378009/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/7"}}},"commit":{"parents":[{"commit":"addbfae6769c1033c618019868314ef3100746d9","subject":"Merge \"Detach volumes when deleting a BFV server pre-scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/addbfae6769c1033c618019868314ef3100746d9"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Shoham Peller","email":"shohamp@gmail.com","date":"2018-02-21 13:46:36.000000000","tz":120},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry.\n\nCo-Authored-By: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d162c1d7160628a3ecd78bb5d5470027168bee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d162c1d7160628a3ecd78bb5d5470027168bee3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"addbfae6769c1033c618019868314ef3100746d9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"dcdd2c9832c7c60fe9163cd744ca2b5acfe16bcc":{"kind":"REWORK","_number":8,"created":"2018-02-22 02:39:38.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/09/378009/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/09/378009/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/09/378009/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/09/378009/8"}}},"commit":{"parents":[{"commit":"addbfae6769c1033c618019868314ef3100746d9","subject":"Merge \"Detach volumes when deleting a BFV server pre-scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/addbfae6769c1033c618019868314ef3100746d9"}]}],"author":{"name":"Shoham Peller","email":"shoham.peller@stratoscale.com","date":"2016-09-27 17:25:05.000000000","tz":180},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-02-22 02:34:50.000000000","tz":-300},"subject":"Handle spawning error on unshelving","message":"Handle spawning error on unshelving\n\nIf spawning fails when unshelving, terminate the volumes\u0027 connections\nwith the node, and remove the node reference from the instance entry.\n\nCo-Authored-By: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\n\nCloses-Bug: 1627694\nChange-Id: I8cfb2280d956d452ccad1fc711bd814b7258147f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcdd2c9832c7c60fe9163cd744ca2b5acfe16bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcdd2c9832c7c60fe9163cd744ca2b5acfe16bcc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"addbfae6769c1033c618019868314ef3100746d9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
