)]}'
{"id":"openstack%2Fnova~38364","triplet_id":"openstack%2Fnova~master~Id02786127ee35124e58425c699e5d3e84f42f527","project":"openstack/nova","branch":"master","topic":"bug/1190845","hashtags":[],"change_id":"Id02786127ee35124e58425c699e5d3e84f42f527","subject":"Remove instance_metadata_get_all* from db api","status":"MERGED","created":"2013-07-23 20:12:13.000000000","updated":"2013-07-23 21:24:32.000000000","submitted":"2013-07-23 21:24:30.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6764629e29cdecb03daecc4c566329c871eef1df","_number":38364,"virtual_id_number":38364,"owner":{"_account_id":6661,"name":"Burt Holzman","email":"holzman@gmail.com","username":"burt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-07-23 21:24:30.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-07-23 20:37:25.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-07-23 20:18:50.000000000","_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":2,"date":"2013-07-23 20:23:38.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":1,"date":"2013-07-23 20:18:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-07-23 20:18:50.000000000","_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":1,"date":"2013-07-23 20:23:38.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-23 20:18:50.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"},{"updated":"2013-07-23 20:18:56.000000000","updated_by":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"reviewer":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"state":"REVIEWER"},{"updated":"2013-07-23 20:23:38.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2013-07-23 20:37:25.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-07-23 21:24:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"63001ff98db4e8cf33db7574fde9c88802cc1cd0","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2013-07-23 20:18:50.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"76a90656e9865618a581618e53ff8fca28e13335","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2013-07-23 20:18:56.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\nseems legit","accounts_in_message":[],"_revision_number":1},{"id":"0abf6489b5b1eec1352cc20ea0e3114e59ab23e4","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2013-07-23 20:23:38.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\nconfirmed to be unused, i\u0027m approving early, assuming in good faith that tests will pass :-)","accounts_in_message":[],"_revision_number":1},{"id":"03e83be82f663c8ec0cf4c4ad3aa8fbce85de5c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-23 20:23:54.000000000","message":"Patch Set 1:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"d01321ef830f41f32e94fee8613d6cb0ed74c681","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-23 20:37:25.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/103418\n\tXenServer (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/103419\n","accounts_in_message":[],"_revision_number":1},{"id":"90bbbd9195935ce1f8e711269d6e13b90d71104d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-23 21:10:28.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/64/38364/1/check/gate-nova-pep8/33941 : SUCCESS in 6m 31s\n- http://docs-draft.openstack.org/64/38364/1/check/gate-nova-docs/28162/doc/build/html/ : SUCCESS in 3m 46s\n- http://logs.openstack.org/64/38364/1/check/gate-nova-python26/29423 : SUCCESS in 21m 33s\n- http://logs.openstack.org/64/38364/1/check/gate-nova-python27/29642 : SUCCESS in 12m 36s\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-full/34863 : SUCCESS in 57m 39s\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-cells-full/2352 : FAILURE in 15m 53s (non-voting)\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-testr-full/1195 : FAILURE in 25m 59s (non-voting)\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-postgres-full/34553 : SUCCESS in 47m 34s\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-neutron/6740 : SUCCESS in 19m 19s (non-voting)\n- http://logs.openstack.org/64/38364/1/check/gate-tempest-devstack-vm-neutron-full/1024 : FAILURE in 46m 53s (non-voting)\n- http://logs.openstack.org/64/38364/1/check/gate-grenade-devstack-vm/20516 : SUCCESS in 27m 46s\n- http://logs.openstack.org/64/38364/1/check/gate-nova-pylint/17702 : SUCCESS in 12m 11s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"55a437984368b6ce58d8e600105c05269b178f8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-23 21:24:30.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/64/38364/1/gate/gate-nova-docs/28165/doc/build/html/ : SUCCESS in 3m 09s\n- http://logs.openstack.org/64/38364/1/gate/gate-nova-pep8/33944 : SUCCESS in 4m 57s\n- http://logs.openstack.org/64/38364/1/gate/gate-nova-python26/29426 : SUCCESS in 25m 03s\n- http://logs.openstack.org/64/38364/1/gate/gate-nova-python27/29645 : SUCCESS in 12m 37s\n- http://logs.openstack.org/64/38364/1/gate/gate-tempest-devstack-vm-full/34868 : SUCCESS in 50m 32s\n- http://logs.openstack.org/64/38364/1/gate/gate-tempest-devstack-vm-postgres-full/34558 : SUCCESS in 47m 29s\n- http://logs.openstack.org/64/38364/1/gate/gate-tempest-devstack-vm-neutron/6746 : SUCCESS in 23m 22s (non-voting)\n- http://logs.openstack.org/64/38364/1/gate/gate-grenade-devstack-vm/20521 : SUCCESS in 26m 54s\n","accounts_in_message":[],"_revision_number":1},{"id":"344dba6012193f16e0a972f61d7dcd80d7cc76ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-23 21:24:32.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"daa26dd561f359ce3ca749ad6f678bf3ea66d574","revisions":{"daa26dd561f359ce3ca749ad6f678bf3ea66d574":{"kind":"REWORK","_number":1,"created":"2013-07-23 20:12:13.000000000","uploader":{"_account_id":6661,"name":"Burt Holzman","email":"holzman@gmail.com","username":"burt"},"ref":"refs/changes/64/38364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/64/38364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/38364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/38364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/64/38364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/64/38364/1"}}},"commit":{"parents":[{"commit":"e20b1274569962347d6e2d42e4d420bf0e7e72e7","subject":"Merge \"Remove project_id from alternate image link path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e20b1274569962347d6e2d42e4d420bf0e7e72e7"}]}],"author":{"name":"Burt Holzman","email":"burt@fnal.gov","date":"2013-07-23 20:08:47.000000000","tz":-300},"committer":{"name":"Burt Holzman","email":"burt@fnal.gov","date":"2013-07-23 20:11:02.000000000","tz":-300},"subject":"Remove instance_metadata_get_all* from db api","message":"Remove instance_metadata_get_all* from db api\n\nThe following calls are no longer called by nova.compute.api and should\nbe removed along with their associated tests:\n\ndb.api.instance_metadata_get_all\ndb.api.sqlalchemy.instance_metadata_get_all\ndb.api.sqlalchemy._instance_metadata_get_all_query\n\nChange-Id: Id02786127ee35124e58425c699e5d3e84f42f527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/daa26dd561f359ce3ca749ad6f678bf3ea66d574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/daa26dd561f359ce3ca749ad6f678bf3ea66d574"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e20b1274569962347d6e2d42e4d420bf0e7e72e7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
