)]}'
{"id":"openstack%2Fnova~392643","triplet_id":"openstack%2Fnova~master~I353afd57be207330757580447a1b606c7b9b7c09","project":"openstack/nova","branch":"master","topic":"bug/1634282","hashtags":[],"change_id":"I353afd57be207330757580447a1b606c7b9b7c09","subject":"Ignore IOError when creating \u0027console.log\u0027","status":"MERGED","created":"2016-11-02 13:09:48.000000000","updated":"2017-04-07 08:20:17.000000000","submitted":"2017-02-01 18:04:00.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"daecb5e1fda932729b0b526c0fc3dceb2cab760d","_number":392643,"virtual_id_number":392643,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-02-01 18:04:00.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2017-01-31 12:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"all":[{"value":1,"date":"2017-01-31 14:45:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":2,"date":"2017-01-31 15:09:46.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-01-31 12:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":1,"date":"2017-02-01 08:34:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":2,"date":"2017-02-01 13:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":1,"date":"2017-02-01 13:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-03 06:24:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2016-11-03 06:43:05.000000000","updated_by":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"reviewer":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"state":"REVIEWER"},{"updated":"2016-11-03 09:23:47.000000000","updated_by":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"reviewer":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"state":"REVIEWER"},{"updated":"2016-11-03 12:15:22.000000000","updated_by":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"reviewer":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"state":"REVIEWER"},{"updated":"2016-11-16 11:56:05.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-06 17:09:36.000000000","updated_by":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"REVIEWER"},{"updated":"2017-01-06 18:36:19.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-01-07 03:05:08.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-11 12:09:48.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-01-11 12:34:56.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-11 13:09:58.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-11 14:50:14.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-11 18:12:00.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-18 11:45:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-31 11:31:48.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-31 12:26:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-31 12:55:18.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2017-01-31 14:22:08.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-01-31 14:45:46.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2017-01-31 15:09:46.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2017-02-01 08:34:08.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2017-02-01 13:32:12.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2017-02-01 18:04:00.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"cbe0e16f7ca10cc523ac6ca69c4b9ee3ae9a0028","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-02 13:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64dad62ef797d151d6a1ec68d46e4f0aea9f34c4","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-11-02 13:17:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0a49c042b7653a8948be273c4d34936a9cdb330","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 13:33:56.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/392643/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"bbb1abb979d38a7d2c7cc106126da4ac4718f299","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-02 13:49:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"060b5475d545691a08f04c10c8517635ebe510aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-02 13:50:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"673d5cea6600625956790bb850a013a4120c5946","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-11-02 13:52:37.000000000","message":"Patch Set 2: Code-Review-1\n\nEven if the fix is dead simple, I guess we still need a tox unit test.","accounts_in_message":[],"_revision_number":2},{"id":"9f281f8f81f71515ae3e3c5865524c98f134cc26","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-02 14:21:07.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/392643/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"bb440cf8ec63369e5baf7623a5e5de778e689da1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 14:58:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/2/check/gate-nova-docs-ubuntu-xenial/5626bd3//doc/build/html/ : SUCCESS in 2m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-nova-pep8-ubuntu-xenial/fddef04/ : FAILURE in 4m 23s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-nova-python27-db-ubuntu-xenial/1eeb460/ : SUCCESS in 8m 38s\n- gate-nova-python34-db http://logs.openstack.org/43/392643/2/check/gate-nova-python34-db/66d6c7c/ : SUCCESS in 7m 03s\n- gate-nova-python35-db-nv http://logs.openstack.org/43/392643/2/check/gate-nova-python35-db-nv/b8b964b/ : SUCCESS in 11m 11s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/995bd72/ : SUCCESS in 44m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4a92efc/ : SUCCESS in 31m 24s\n- gate-tempest-dsvm-full-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-full-ubuntu-xenial/0929868/ : SUCCESS in 38m 43s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-grenade-dsvm-ubuntu-xenial/177f271/ : SUCCESS in 24m 05s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-nova-tox-db-functional-ubuntu-xenial/b4b3ae4/ : SUCCESS in 13m 46s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/43/392643/2/check/gate-grenade-dsvm-neutron-multinode/ae02df8/ : SUCCESS in 35m 28s\n- gate-grenade-dsvm-multinode-live-migration-nv http://logs.openstack.org/43/392643/2/check/gate-grenade-dsvm-multinode-live-migration-nv/c5b5e1b/ : SUCCESS in 31m 39s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/8792473/ : SUCCESS in 28m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/3f65cd1/ : SUCCESS in 41m 43s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/7e48a53/ : SUCCESS in 48m 42s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/4d2a825/ : SUCCESS in 1h 06m 27s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/db315ec/ : SUCCESS in 18m 44s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/43/392643/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/63457a4/ : SUCCESS in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"132f0a51174b6b16352c25fa519e57ed7795d00e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-02 15:22:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/43/392643/2/check/dsvm-tempest-xen/4423cd2 : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8e97e1b55b942f902434dc55082da80943de1790","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 15:44:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/2/check/dsvm-tempest-kvm/82f4dc7 : SUCCESS in 1h 03m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"30470364c64a33aab83e56c83331e67697c78667","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-02 18:55:38.000000000","message":"Patch Set 2: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/392643/2/1461/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/392643/2/1461/ : SUCCESS (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":2},{"id":"56ec6dfca38ea0f7aa4f9d02cff371f395a98fde","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 19:46:57.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/2/ : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":2},{"id":"9b50597eda7db0c7d02a133cb834b4ed304c8cf5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2016-11-02 20:19:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/43/392643/2/check-nova/Nova-ML2-Sriov/139baef : SUCCESS in 33m 18s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/43/392643/2/check-nova/Nova-MACVTAP-ML2-Sriov/eeb18cb : SUCCESS in 34m 13s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"9237a87a09b07d0e9e4b1be1001b61cc975751a3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-02 21:18:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/2/check/tempest-dsvm-full-nfv-xenial/95b9b1f : FAILURE in 27m 24s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/2/check/tempest-dsvm-intel-nfv-xenial/6f2840c : FAILURE in 30m 25s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/55bca57 : FAILURE in 33m 59s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/b4031f0 : SUCCESS in 1h 10m 16s (non-voting)\n- tempest-dsbaremetal-pcipassthrough-xenial http://intel-openstack-ci-logs.ovh/43/392643/2/check/tempest-dsbaremetal-pcipassthrough-xenial/5b1062b : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b105ff2603317f7e7c8b6cde490836e7de4a5d98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2016-11-03 06:24:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdeadaea3733fe04a0f66a6de235fea83b0c7c9b","author":{"_account_id":20191,"name":"Dinesh Bhor","email":"bhordinesh07@gmail.com","username":"dineshbhor"},"date":"2016-11-03 06:43:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"853745bf9532be5c2fb487304f02d10cdb8b66a8","author":{"_account_id":21784,"name":"Radek Zetik","email":"radekx.zetik@intel.com","username":"rzetikx"},"date":"2016-11-03 09:23:47.000000000","message":"Patch Set 2:\n\nintel-nfv-ci recheck","accounts_in_message":[],"_revision_number":2},{"id":"4af8d490b7d12253822a6b2afe27f3ac94da8cac","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2016-11-03 11:28:00.000000000","message":"Patch Set 2:\n\nrun nova-quobyteci-dsvm-volume","accounts_in_message":[],"_revision_number":2},{"id":"57cf094bdb61d30dd2cc34f14dfe0e4fba507057","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-03 11:39:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bddd73fbc15fe0a9e585f48f7b2ab584c87b6ae0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-03 12:01:57.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/392643/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"46d85a3ca029605b53d59706e3b09a537fb740c2","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-11-03 12:15:22.000000000","message":"Patch Set 3:\n\nLooks good.\n\nStill missing a unit test.","accounts_in_message":[],"_revision_number":3},{"id":"a317c197f5fee5fcfcc4c9dd9f36b2186a277eec","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-11-03 12:15:32.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"927e6880aa7edf67788a8b749bca52ef38b94d8b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-03 12:18:08.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"1d1df5a4fd5954aebc0cdd34a6945bcebfc3e7fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-03 12:18:59.000000000","message":"Patch Set 3:\n\n\u003e Looks good.\n \u003e \n \u003e Still missing a unit test.\n\nI\u0027m not sure if we need unit tests here - there are no existing tests for this function as it stands :/","accounts_in_message":[],"_revision_number":3},{"id":"5887ab6ac0a167b922f3c8c9dd153b1bba6895fc","author":{"_account_id":17973,"name":"Ludovic Beliveau","email":"lbeliveau@gmail.com","username":"lbeliveau"},"date":"2016-11-03 12:20:40.000000000","message":"Patch Set 3: -Code-Review\n\nHopefully cores will not ask for it.","accounts_in_message":[],"_revision_number":3},{"id":"453ccda3c20c91528b3ceeb01cff22a53552c541","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 12:27:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/3/check/check-dsvm-tempest-vz7-exe-minimal/cbb2b61 : SUCCESS in 47m 35s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"68ff87dda251edf5f3b1e9240f86b84c14b904d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-03 12:34:12.000000000","message":"Patch Set 3: Verified+1\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/392643/3/1515/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/392643/3/1515/ : SUCCESS (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"26113592044096ee43fd8fce737eb8ab7f2b885d","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 12:40:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/3/check/dsvm-tempest-kvm/141661b : SUCCESS in 1h 00m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027Virtuozzo Storage CI recheck\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"40962f0dc68d1e22516ee90688af86b3e9d6155a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 12:45:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/3/check/gate-nova-docs-ubuntu-xenial/b40cb5c//doc/build/html/ : SUCCESS in 6m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-pep8-ubuntu-xenial/d99456f/ : SUCCESS in 7m 42s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-python27-db-ubuntu-xenial/571f8cd/ : SUCCESS in 14m 08s\n- gate-nova-python34-db http://logs.openstack.org/43/392643/3/check/gate-nova-python34-db/3a66473/ : SUCCESS in 9m 35s\n- gate-nova-python35-db-nv http://logs.openstack.org/43/392643/3/check/gate-nova-python35-db-nv/3a4ea95/ : SUCCESS in 8m 34s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6750f7c/ : SUCCESS in 44m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/36a1c01/ : SUCCESS in 50m 58s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-ubuntu-xenial/15ffd1c/ : SUCCESS in 27m 42s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-tox-db-functional-ubuntu-xenial/9910e4d/ : SUCCESS in 18m 24s\n- gate-grenade-dsvm-neutron-multinode http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-multinode/27b0d54/ : SUCCESS in 34m 40s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/97a1a59/ : SUCCESS in 32m 44s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/9d5b796/ : FAILURE in 29m 36s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7ae9118/ : SUCCESS in 56m 38s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/5433c2d/ : FAILURE in 53m 56s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/3787fdb/ : SUCCESS in 1h 00m 04s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8d67cd9/ : SUCCESS in 22m 34s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ssh-tinyipa-nv/6829d15/ : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6bdfb4405028b9c2be6f07c766a923df04486a4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2016-11-03 12:54:01.000000000","message":"Patch Set 3:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/3/check/check-ibm-tempest-dsvm-full/9daf3da/ : SUCCESS in 57m 34s (non-voting)\n- check-ibm-tempest-dsvm-neutron-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/3/check/check-ibm-tempest-dsvm-neutron-full/963b44f/ : SUCCESS in 1h 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15ea27ea70376baa6d418317529f6e7fe438ea8f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2016-11-03 13:12:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/43/392643/3/check/dsvm-tempest-xen/22761c1 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":3},{"id":"98eeee681ae9172a8b4c35fdf6ed0716b1abb690","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-03 15:40:09.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/3/ : FAILURE in 4h 00m 02s","accounts_in_message":[],"_revision_number":3},{"id":"46c33dc7cb64fec6e2310728a6853f351476c589","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-04 01:45:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-full-nfv-xenial/bbe182f : SUCCESS in 1h 56m 04s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-intel-nfv-xenial/49ee2bd : SUCCESS in 39m 46s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/13ecd93 : SUCCESS in 54m 00s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/4f6b13c : SUCCESS in 1h 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"599f42e40a5fee8165c8b026fc9984c877e81003","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2016-11-11 14:06:24.000000000","message":"Patch Set 3:\n\nrun nova-quobyteci-dsvm-volume","accounts_in_message":[],"_revision_number":3},{"id":"ef7d61a283a416174ec482980ff89cec05104d45","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-11 15:03:35.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"e374832c634f0fd46fa4c6ecd1787804cae4e04b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-11-16 11:22:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"27c511a1f805932be4d0f1236a82b5ffb8c26050","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-16 11:56:05.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/3/ : FAILURE in 32m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2775a69bf0be334ea0b87d0bb3cee01cca52ddd2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-16 12:54:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/3/check/gate-nova-docs-ubuntu-xenial/720a979//doc/build/html/ : SUCCESS in 2m 39s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-pep8-ubuntu-xenial/61193cc/ : SUCCESS in 5m 12s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-python27-db-ubuntu-xenial/c585cb7/ : SUCCESS in 9m 06s\n- gate-nova-python34-db http://logs.openstack.org/43/392643/3/check/gate-nova-python34-db/ef91435/ : SUCCESS in 7m 58s\n- gate-nova-python35-db-nv http://logs.openstack.org/43/392643/3/check/gate-nova-python35-db-nv/82a6451/ : SUCCESS in 8m 15s (non-voting)\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c194783/ : SUCCESS in 49m 03s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/22dbe0b/ : SUCCESS in 33m 45s\n- gate-grenade-dsvm-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-ubuntu-xenial/2fd0be1/ : SUCCESS in 27m 01s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-nova-tox-db-functional-ubuntu-xenial/7361e4f/ : SUCCESS in 16m 14s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8d59a37/ : SUCCESS in 1h 01m 50s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/3/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/8106fb7/ : POST_FAILURE in 36m 01s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/ad04749/ : SUCCESS in 43m 38s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b0e69a2/ : SUCCESS in 1h 13m 40s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial/f9fac8c/ : SUCCESS in 54m 57s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/042480c/ : SUCCESS in 1h 30m 26s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/ecf1364/ : SUCCESS in 25m 07s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/0d83184/ : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dceed84cc701311f2a7454b85dba38208329931","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2016-11-16 14:27:29.000000000","message":"Patch Set 3:\n\n- ext-nova-dsvm http://208.91.1.172/logs/ext-nova-dsvm/392643/3/2199/ : SUCCESS\n- ext-nova-dsvm-nsx http://208.91.1.172/logs/ext-nova-dsvm-nsx/392643/3/2199/ : SUCCESS (non-voting)\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":3},{"id":"cae149a1f2f2078152cc2a15e6c65ae3bb7c7ef4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-16 15:15:00.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"1f306093a5f0d7c7735faff340c37bb93943e26d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-16 21:28:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-full-nfv-xenial/720e8ef : SUCCESS in 2h 36m 50s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-intel-nfv-xenial/7088563 : SUCCESS in 41m 00s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/2aa14bd : SUCCESS in 1h 34m 31s\n- tempest-dsbaremetal-pcipassthrough-xenial http://intel-openstack-ci-logs.ovh/43/392643/3/check/tempest-dsbaremetal-pcipassthrough-xenial/c48a772 : FAILURE in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2e714d169c9f0726453f3919a006e9c24922a320","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2016-11-17 12:42:48.000000000","message":"Patch Set 3:\n\nrun nova-quobyteci-dsvm-volume","accounts_in_message":[],"_revision_number":3},{"id":"e7ff91bf70a9c5908b7cfd9ca8f507708342a45d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2016-11-18 01:50:19.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"ed62de02e2b69178928952766c8b40fdfb8fe042","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2017-01-06 16:52:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nstill missing unittest coverage","accounts_in_message":[],"_revision_number":3},{"id":"769e2d877962775ccf0db49d7c3b70c6ddaa2565","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-06 17:06:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"118af0064315d2e9a0710a4841e1f9ee5bead917","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-06 17:06:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ff2d73f372591affb0582d2e48aafe73779e387","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2017-01-06 17:09:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e31cb89a4017577704f1eaebd148f05dda082da9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-06 17:32:46.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/392643/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"84aeb6a806ce495c0f28b8b793dfc220dddd7c5a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-06 17:51:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/43/392643/4/check-nova/Nova-ML2-Sriov/ed13cb7 : SUCCESS in 42m 59s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/43/392643/4/check-nova/Nova-MACVTAP-ML2-Sriov/dc2abc3 : SUCCESS in 41m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"2717b9c0fc88704f0ec552a3f78629e3ed8b4c18","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-06 18:18:13.000000000","message":"Patch Set 4: Verified+1\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/tempest-dsvm-full-xenial/16669a2/ : SUCCESS in 1h 11m 08s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/grenade-dsvm-xenial/57d9d6a/ : SUCCESS in 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1633e8f1dc0550610d4190b253c405057f700cd4","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-06 18:23:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/4/ : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":4},{"id":"4b819a8df66742659f638f2f0e37a67f24f22736","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-06 18:28:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/4/check/gate-nova-docs-ubuntu-xenial/e950dee//doc/build/html/ : SUCCESS in 2m 50s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-pep8-ubuntu-xenial/5e0d3fd/ : SUCCESS in 3m 45s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-python27-db-ubuntu-xenial/02f9d08/ : SUCCESS in 15m 40s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/4/check/gate-nova-python35-db/7a4b59d/ : FAILURE in 8m 17s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/53fbaa7/ : SUCCESS in 1h 20m 38s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/9476dec/ : SUCCESS in 59m 01s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-tox-db-functional-ubuntu-xenial/5b7419a/ : SUCCESS in 13m 11s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/6d9d402/ : SUCCESS in 38m 04s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/b4476f1/ : FAILURE in 42m 59s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/7796f18/ : SUCCESS in 38m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/349a637/ : SUCCESS in 1h 01m 26s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/bdd37bf/ : SUCCESS in 1h 00m 14s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/11c1641/ : SUCCESS in 1h 16m 07s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6134895/ : SUCCESS in 24m 06s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ae158ba/ : SUCCESS in 38m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05ce9a2d74d0bd9a9f26bd206b5820a7c7703f06","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-06 18:36:19.000000000","message":"Patch Set 4:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python35-db: https://bugs.launchpad.net/bugs/1654393\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"12447581597f93babbca8731d24437490b34b5b5","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-06 21:21:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-kvm/04077d1 : SUCCESS in 56m 29s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-exe/a5607f6 : SUCCESS in 45m 55s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"fab3e8804390454584755843149a53a0025902f0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-01-07 03:05:08.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/43/392643/4/check/dsvm-tempest-xen/943f60f : FAILURE in 2h 46m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5130064437ac4d7b88d4d3314d4fcf9d79917333","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-07 06:51:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-full-nfv-xenial/1dfdfe8 : FAILURE in 2h 16m 54s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-intel-nfv-xenial/f2783f7 : SUCCESS in 42m 41s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9f040c2 : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":4},{"id":"372498a79b8cbe75f4a79b638bfc443a8dbf4cf1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-08 20:46:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/4/check/check-dsvm-tempest-vz7-exe-minimal/a23597a : SUCCESS in 1h 37m 00s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/4/check/check-dsvm-tempest-vz7-exe-multinode/9c75cd2 : SUCCESS in 53m 03s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027recheck Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"1fe3db342604baeb275a4c8e7aefc6869aab974c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-09 09:51:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6afba103220c1ab101ad8bbb8a1fc89cd2da5e17","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-09 10:57:28.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/tempest-dsvm-full-xenial/76da2d8/ : SUCCESS in 1h 02m 30s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/grenade-dsvm-xenial/54344fe/ : SUCCESS in 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"21fd61aa35e95d88cdc39ecfb320dfd6997c6298","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-09 11:15:03.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/4/ : FAILURE in 1h 20m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ef0d67748ab64b031bf0e039920b87d2cfb367f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-09 12:01:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/4/check/gate-nova-docs-ubuntu-xenial/cc516d7//doc/build/html/ : SUCCESS in 5m 00s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-pep8-ubuntu-xenial/a358579/ : SUCCESS in 3m 31s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-python27-db-ubuntu-xenial/905acf6/ : SUCCESS in 7m 55s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/4/check/gate-nova-python35-db/eb45b95/ : SUCCESS in 7m 40s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/76c46e3/ : SUCCESS in 58m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/653c10a/ : SUCCESS in 33m 50s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-tox-db-functional-ubuntu-xenial/223a69c/ : SUCCESS in 10m 19s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/7978b5a/ : SUCCESS in 37m 52s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a85e6af/ : FAILURE in 38m 42s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/2bca6e7/ : SUCCESS in 34m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e5f0ce9/ : FAILURE in 55m 58s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/83336a7/ : SUCCESS in 2h 03m 52s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/90cd255/ : SUCCESS in 1h 02m 34s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/dc76996/ : SUCCESS in 37m 47s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/2aa4e0b/ : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"02b53593a9bf8ceef17441e6622555abd0cb302e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-09 12:01:25.000000000","message":"Patch Set 4:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial: https://bugs.launchpad.net/bugs/1654762 https://bugs.launchpad.net/bugs/1449136\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"df3c427c9b72013115598c33d90e60d128b6fc7c","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-09 13:50:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-kvm/8e3ef63 : FAILURE in 5m 45s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-exe/6a1d2c2 : FAILURE in 9m 27s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"95a76de15263fe56818ebb588af6566a914aaa57","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-09 13:58:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-full-nfv-xenial/835f1fd : SUCCESS in 1h 54m 08s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-intel-nfv-xenial/600886a : SUCCESS in 41m 27s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/43/392643/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/816ba8e : SUCCESS in 1h 42m 13s","accounts_in_message":[],"_revision_number":4},{"id":"798bfe2c1a5b35c5e6afec7eeb546128f8da2e66","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-11 10:51:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5d8eb8cbdfc5264a174fed01def4b1007e20d5aa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-11 11:50:30.000000000","message":"Patch Set 4:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/tempest-dsvm-full-xenial/bc603a3/ : SUCCESS in 58m 43s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/4/check/grenade-dsvm-xenial/b9e938e/ : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f03c8be891f210108d94c6aec432d9e05cb3621e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-11 11:59:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/4/check/gate-nova-docs-ubuntu-xenial/5117e5c//doc/build/html/ : SUCCESS in 4m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-pep8-ubuntu-xenial/8fd7784/ : SUCCESS in 3m 48s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-python27-db-ubuntu-xenial/16145f9/ : SUCCESS in 7m 23s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/4/check/gate-nova-python35-db/3c4eea0/ : SUCCESS in 7m 43s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/cb949c5/ : SUCCESS in 1h 01m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8b1f087/ : SUCCESS in 32m 40s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-tox-db-functional-ubuntu-xenial/062863a/ : SUCCESS in 11m 04s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/9fcf6b3/ : FAILURE in 7m 47s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/bf73a84/ : SUCCESS in 46m 44s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a987236/ : SUCCESS in 1h 01m 49s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/19904bd/ : SUCCESS in 36m 14s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/6a85f4e/ : SUCCESS in 1h 05m 55s\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/49a1d6f/ : SUCCESS in 58m 22s (non-voting)\n- gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-neutron-placement-full-ubuntu-xenial-nv/fa38c4d/ : SUCCESS in 51m 54s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/edd2ba4/ : SUCCESS in 27m 38s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/517b691/ : SUCCESS in 43m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"caa160008513ed6524662de746811f1539a3dad9","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-11 12:09:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/4/ : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ddf9cf3390ab9087d138db84d1fb075b5bf4fed6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-11 12:31:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/4/check/check-dsvm-tempest-vz7-exe-minimal/ae13b4b : SUCCESS in 58m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"bbc8f03c96312a92a5f7964c7614cde342d0fbe8","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-11 12:34:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-kvm/5c11490 : FAILURE in 1h 03m 13s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/4/check/dsvm-tempest-exe/6254e5e : SUCCESS in 46m 19s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"bc8b84958ef4e08cac38d26611fb864cbaed981c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-11 13:09:58.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"f2c1ed1a5f5b78330789523b7384375ac2694cdc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-11 14:50:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-11/392643/4/check/tempest-dsvm-full-nfv-xenial/4ba5c4a : UNSTABLE in 1h 59m 00s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-11/392643/4/check/tempest-dsvm-intel-nfv-xenial/49b0794 : SUCCESS in 42m 19s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-11/392643/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/2612a64 : SUCCESS in 1h 44m 27s","accounts_in_message":[],"_revision_number":4},{"id":"cbc608d6c72d8e755110d7b52536f1c1464b2549","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-11 18:12:00.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/4/check-migration/check-dsvm-tempest-vz7-exe-multinode/206616e : FAILURE in 39m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-migration-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"e015b0aa7fd705cb81fc11cce0a2a00f190002ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-14 05:24:57.000000000","message":"Patch Set 4:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/392643/4/1858/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":4},{"id":"b0a7de6e297d2d923ebb990ade302d334b784b6b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-01-18 11:45:30.000000000","message":"Patch Set 4:\n\n- ext-nova-dsvm-nsx http://207.189.188.190/logs/ext-nova-dsvm-nsx/392643/4/1997/ : FAILURE\n\n\n\nAdd review comment \"vmware-recheck-patch\" to retrigger a new build.","accounts_in_message":[],"_revision_number":4},{"id":"346747166a32662a48432f772f2e5673e59e6b65","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-01-23 11:11:30.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9fd0ffb19ddaa079ed890de34c23c3c5efc01697","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2017-01-30 18:46:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nInquiries within :)","accounts_in_message":[],"_revision_number":4},{"id":"7f077ee90843113d74140ffd0dacedc85d04caf2","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-01-31 10:26:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f5023330e509bac8376216fde1409a625595d4ca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-31 11:08:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0efecb5f9c2c0363b2d07b5b60aab8e717b04e96","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-31 11:17:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/43/392643/5/check-nova/Nova-ML2-Sriov/af1cd35 : FAILURE in 4m 03s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/43/392643/5/check-nova/Nova-MACVTAP-ML2-Sriov/1999fe6 : FAILURE in 1m 00s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"509fe358122d9e0b293b555b1baf959928426cc1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-31 11:24:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d639c71060a38b98995526d99d73ce85d5ce4e0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-31 11:31:48.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/392643/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"09dce815d7589dfd435014ef31b013ca4d860289","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-01-31 11:51:32.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/392643/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"0158124714fe8543f9f683fca0cdd990afe4eecc","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-31 12:23:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/6/check/check-dsvm-tempest-vz7-exe-minimal/3d74df4 : SUCCESS in 58m 54s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/6/check/check-dsvm-tempest-vz7-exe-multinode/c458e61 : SUCCESS in 55m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"75070ee7521d0a4f319d10bc291923c334bfba3d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-31 12:26:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/6/check/tempest-dsvm-full-xenial/6b85171/ : SUCCESS in 1h 01m 11s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/6/check/grenade-dsvm-xenial/ebafecc/ : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"30211d9ca7f6e554d620ec4855c0a202cc48118d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-01-31 12:38:27.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027recheck powervm\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/43/392643/6/check/nova-out-of-tree-pvm/9cadd35 : FAILURE in 1h 13m 35s","accounts_in_message":[],"_revision_number":6},{"id":"92b9b386f891612a67149f6b45e64dbe34421392","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-01-31 12:55:18.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks for the update.","accounts_in_message":[],"_revision_number":6},{"id":"57bb683468408a420dee1def181229011dfadbd6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-31 13:08:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/6/check/gate-nova-docs-ubuntu-xenial/ddabfa6//doc/build/html/ : SUCCESS in 3m 33s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-pep8-ubuntu-xenial/669d546/ : SUCCESS in 5m 52s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-python27-db-ubuntu-xenial/cc26cfd/ : SUCCESS in 12m 36s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/6/check/gate-nova-python35-db/4a59a7f/ : SUCCESS in 10m 26s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d6b0027/ : SUCCESS in 55m 05s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/0de24e3/ : SUCCESS in 33m 26s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-tox-db-functional-ubuntu-xenial/3c3f4d9/ : SUCCESS in 10m 25s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/b7eac25/ : SUCCESS in 10m 59s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a800b35/ : SUCCESS in 40m 25s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/dd246c9/ : FAILURE in 35m 08s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/3f067b2/ : SUCCESS in 1h 26m 51s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-cells-ubuntu-xenial/5b1b613/ : FAILURE in 41m 13s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/fb30dfa/ : FAILURE in 1h 03m 44s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/4ff1086/ : FAILURE in 1h 41m 03s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/f94253f/ : SUCCESS in 24m 16s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c187ebf/ : SUCCESS in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c2dc5a5a54036de5b5d29dc6de91248eab58944d","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-31 13:09:53.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/6/ : FAILURE in 1h 44m 59s","accounts_in_message":[],"_revision_number":6},{"id":"3d4358d84290133d8bba129812dbcc77e6400fcc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-01-31 13:27:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5c53420f40335d2293f3bc58bbf825a74d35d8a9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-01-31 13:28:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-31/392643/6/check/tempest-dsvm-full-nfv-xenial/1fb255f : SUCCESS in 1h 58m 37s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-01-31/392643/6/check/tempest-dsvm-intel-nfv-xenial/30dbcd0 : FAILURE in 52m 02s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-01-31/392643/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/4a650ba : SUCCESS in 1h 59m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d51df2d8337adfaebbeaf247057c5e0a0dfe11af","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-01-31 13:42:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/6/check/dsvm-tempest-kvm/f8804a0 : SUCCESS in 1h 01m 51s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/43/392643/6/check/dsvm-tempest-exe/9ac0f3e : SUCCESS in 44m 33s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"9829c34dcda05c73e247e2062f03e80d0f79367d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-01-31 14:22:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/43/392643/5/check-nova/Nova-ML2-Sriov/6aa58e3 : SUCCESS in 48m 34s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/43/392643/5/check-nova/Nova-MACVTAP-ML2-Sriov/e369aa6 : SUCCESS in 48m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"0010b66e27cd73ab2291469b57008d1326c6316e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-01-31 14:32:53.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/6/check/check-dsvm-tempest-vz7-exe-minimal/defa10f : SUCCESS in 1h 04m 41s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/43/392643/6/check/check-dsvm-tempest-vz7-exe-multinode/af3cf22 : FAILURE in 53m 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"df16f83215863c94c3dbe72b056411de68a25ae2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-01-31 14:35:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/6/check/tempest-dsvm-full-xenial/47d6462/ : SUCCESS in 1h 06m 12s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/43/392643/6/check/grenade-dsvm-xenial/9c4588e/ : SUCCESS in 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ef4c8baa0e550d783b9bc00bdba141d23beaed41","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2017-01-31 14:45:46.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d05f03a9d480cb2c8c96dcd1edd87f7e39baa17b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-01-31 14:57:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/392643/6/ : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":6},{"id":"6752b20bd33ef9d2afb3872f725d8de72bb19050","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-31 15:04:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/6/check/gate-nova-docs-ubuntu-xenial/421cd13//doc/build/html/ : SUCCESS in 2m 26s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-pep8-ubuntu-xenial/c9cd2cc/ : SUCCESS in 6m 06s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-python27-db-ubuntu-xenial/b5c70d1/ : SUCCESS in 7m 22s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/6/check/gate-nova-python35-db/1c294ed/ : SUCCESS in 7m 30s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/caf00a9/ : SUCCESS in 1h 01m 50s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e738c60/ : SUCCESS in 32m 22s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-tox-db-functional-ubuntu-xenial/1facc34/ : SUCCESS in 10m 19s\n- gate-nova-tox-db-functional-py35-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-nova-tox-db-functional-py35-ubuntu-xenial/3997b4f/ : SUCCESS in 10m 46s (non-voting)\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/dff6445/ : SUCCESS in 58m 09s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/43/392643/6/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/140d38b/ : FAILURE in 54m 56s (non-voting)\n- gate-tempest-dsvm-py35-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-py35-ubuntu-xenial-nv/5f6c08e/ : FAILURE in 1h 11m 57s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-cells-ubuntu-xenial/7e090f1/ : SUCCESS in 36m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/e1eb9cd/ : FAILURE in 1h 00m 47s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/c49ad63/ : SUCCESS in 1h 32m 19s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a3d1c2f/ : SUCCESS in 36m 51s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/43/392643/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-agent_ipmitool-tinyipa-ubuntu-xenial-nv/3f4b1e4/ : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cb13d86a8f501d817b4708e3d6bb78493636d73e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2017-01-31 15:09:46.000000000","message":"Patch Set 6: Code-Review+2\n\nThat feels much better to me, checking for the permissions error.","accounts_in_message":[],"_revision_number":6},{"id":"d91b2ac7fb6bb1346ec1fb462d2779c7c5bd8f74","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-01-31 16:00:26.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-43-392643-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"a02186d94ba68233b5e2074c0c865c5c19fc3499","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-31 17:54:15.000000000","message":"Patch Set 6:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-cells-ubuntu-xenial: https://bugs.launchpad.net/bugs/1660511\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"38523f75b04c3410156ca840e733a7db98a3e2b2","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2017-02-01 08:34:08.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a6931245030226082362b72d4d79a7adb8fab813","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-02-01 13:32:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3502508944edda021751899264f69edd4be9e63c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-01 13:32:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"5b1ccc8c7945acc276b4859c9d9503339e2c17d6","date":"2017-02-01 18:04:00.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":6},{"id":"f2059199e7cf9fc9e9f4b30b72897dc70da7c254","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-01 18:04:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/43/392643/6/gate/gate-nova-docs-ubuntu-xenial/8ef6881//doc/build/html/ : SUCCESS in 4m 36s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-nova-pep8-ubuntu-xenial/c438c3b/ : SUCCESS in 6m 19s\n- gate-nova-python27-db-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-nova-python27-db-ubuntu-xenial/2f7065d/ : SUCCESS in 11m 31s\n- gate-nova-python35-db http://logs.openstack.org/43/392643/6/gate/gate-nova-python35-db/be05db2/ : SUCCESS in 7m 37s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/10912cd/ : SUCCESS in 1h 21m 48s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/82918c4/ : SUCCESS in 32m 39s\n- gate-nova-tox-db-functional-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-nova-tox-db-functional-ubuntu-xenial/a7f62fb/ : SUCCESS in 10m 55s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-tempest-dsvm-cells-ubuntu-xenial/acb310d/ : SUCCESS in 1h 01m 12s\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/43/392643/6/gate/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0af1b27/ : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":6},{"id":"16e8dea24659268d600efcaf419d1ae5cb67caac","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-04-07 08:20:17.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/newton as commit a96092d0e48fef62d257acf0428ddb6daf07d3a4","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1848860fd4fb6dfb2ead1f8271050d36df592205","revisions":{"7063d143bab0ef8cec80b798e555889e3c6cde14":{"kind":"REWORK","_number":1,"created":"2016-11-02 13:09:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/1"}}},"commit":{"parents":[{"commit":"9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb","subject":"Merge \"api-ref: document the power_state enum values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 13:08:30.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7063d143bab0ef8cec80b798e555889e3c6cde14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7063d143bab0ef8cec80b798e555889e3c6cde14"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0a3d3a6bd8c6b51fbdd7689177da66d4a253ccee":{"kind":"REWORK","_number":2,"created":"2016-11-02 13:50:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/2"}}},"commit":{"parents":[{"commit":"9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb","subject":"Merge \"api-ref: document the power_state enum values\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 13:49:57.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a3d3a6bd8c6b51fbdd7689177da66d4a253ccee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a3d3a6bd8c6b51fbdd7689177da66d4a253ccee"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9086ea3ad5a3cf38d7b908cbce0b21d3c876cadb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5c9c159f18de0fcddbfcb35734dbb14f557df6b6":{"kind":"REWORK","_number":3,"created":"2016-11-03 11:39:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/3"}}},"commit":{"parents":[{"commit":"8ecef74b276d3ac92a568747be174c22f3c4be4a","subject":"Merge \"Remove conductor local api:s and \u0027use_local\u0027 config option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ecef74b276d3ac92a568747be174c22f3c4be4a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-03 10:39:41.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c9c159f18de0fcddbfcb35734dbb14f557df6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c9c159f18de0fcddbfcb35734dbb14f557df6b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8ecef74b276d3ac92a568747be174c22f3c4be4a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0ff2960257af265cd6a46fa73bcf4a8cab2d0254":{"kind":"REWORK","_number":4,"created":"2017-01-06 17:06:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/4"}}},"commit":{"parents":[{"commit":"45f14c2018ea4f857ab380cdf11f8da0ad2cec0a","subject":"Merge \"conf: remove deprecated barbican options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/45f14c2018ea4f857ab380cdf11f8da0ad2cec0a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-01-06 17:05:47.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ff2960257af265cd6a46fa73bcf4a8cab2d0254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ff2960257af265cd6a46fa73bcf4a8cab2d0254"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"45f14c2018ea4f857ab380cdf11f8da0ad2cec0a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2085418d5b92bd4a346baabbf30860ab34353b71":{"kind":"REWORK","_number":5,"created":"2017-01-31 11:08:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/5"}}},"commit":{"parents":[{"commit":"dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2","subject":"Merge \"libvirt: Use the mirror element to detect job completion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-01-31 11:08:11.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2085418d5b92bd4a346baabbf30860ab34353b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2085418d5b92bd4a346baabbf30860ab34353b71"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1848860fd4fb6dfb2ead1f8271050d36df592205":{"kind":"REWORK","_number":6,"created":"2017-01-31 11:24:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/392643/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/392643/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/392643/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/392643/6"}}},"commit":{"parents":[{"commit":"dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2","subject":"Merge \"libvirt: Use the mirror element to detect job completion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2016-11-02 12:43:35.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-01-31 11:23:54.000000000","tz":0},"subject":"Ignore IOError when creating \u0027console.log\u0027","message":"Ignore IOError when creating \u0027console.log\u0027\n\nIn \u0027d4e6bd8\u0027, and later refined in \u0027ec6ed24c\u0027, an issue with the the\nQuobyte CI was resolved by ensuring that the \u0027console.log\u0027 file is\nalways created, even when libvirt\u0027s \u0027dynamic_ownership\u0027 config option\nis set to \u00270\u0027. However, attempting to take ownership of this file can\ncause issues when \u0027dynamic_ownership\u0027 is set to \u00271\u0027 and a user attempts\nto reschedule an instance to a host where it was previously located. In\nthis case, the permissions on the file have already been co-opted by\nlibvirt, and attempting to change them results in an IOError.\n\nThe easiest fix for this is to simply ignore the IOError, as the file\nalready exists and libvirt has set correct permissions already. This is\nwhat do here. A longer term fix would be to avoid changing the\npermissions only for the offending Quobytes block devices, but this is\nlikely not backportable and will be done in a follow-up fix.\n\nChange-Id: I353afd57be207330757580447a1b606c7b9b7c09\nPartial-bug: #1634282\nRelated-bug: #1597644\nRelated-bug: #1609298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1848860fd4fb6dfb2ead1f8271050d36df592205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1848860fd4fb6dfb2ead1f8271050d36df592205"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"dae6b760b9c40bbf3b72a0218dbf1dbc823f30e2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
