)]}'
{"id":"openstack%2Fnova~402535","triplet_id":"openstack%2Fnova~master~I2b02fc2968993947bec7e3b0e878b0ea149a9b31","project":"openstack/nova","branch":"master","topic":"project-badges","hashtags":[],"change_id":"I2b02fc2968993947bec7e3b0e878b0ea149a9b31","subject":"Show team and repo badges on README","status":"MERGED","created":"2016-11-25 12:58:26.000000000","updated":"2016-12-28 12:33:33.000000000","submitted":"2016-12-28 12:33:32.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b2ba39154278af66dd8f523ac52e8422f17e2f8","_number":402535,"virtual_id_number":402535,"owner":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},{"value":0,"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},{"value":2,"date":"2016-12-28 12:33:32.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2016-11-25 18:34:26.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},"all":[{"value":1,"date":"2016-12-27 10:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},{"value":1,"date":"2016-12-05 22:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-22 18:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2016-12-28 09:44:27.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2016-11-27 02:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},{"value":0,"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2016-12-28 09:44:27.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},{"value":0,"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},{"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-11-25 18:34:26.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2016-11-27 02:07:04.000000000","updated_by":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"reviewer":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"state":"REVIEWER"},{"updated":"2016-12-05 22:43:34.000000000","updated_by":{"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},"reviewer":{"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},"state":"REVIEWER"},{"updated":"2016-12-22 18:04:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2016-12-27 10:12:26.000000000","updated_by":{"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},"reviewer":{"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},"state":"REVIEWER"},{"updated":"2016-12-28 09:44:27.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2016-12-28 12:33:32.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1dca8ccf1caec4e0c5f1357e9fea6539b5b4cf8f","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-11-25 12:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"256ed153b18df7ed63acc3b00a1ff72a80d87041","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-25 13:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/35/402535/1/check/gate-nova-docs-ubuntu-xenial/9ce7999//doc/build/html/ : SUCCESS in 2m 47s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/35/402535/1/check/gate-nova-pep8-ubuntu-xenial/c1c50f0/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"94c898931cc7ba15574b9a785f9e4277239897a6","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2016-11-25 14:22:33.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/402535/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f595f575ff6905020174f3c2d63cc85f8f7a1d3c","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2016-11-25 17:18:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/402535/1/ : SUCCESS in 2h 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f3cc219d7080480227acf9e372f87b63e5589559","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2016-11-25 18:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/35/402535/1/check/tempest-dsvm-full-nfv-xenial/4b4d434 : SUCCESS in 2h 26m 52s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/35/402535/1/check/tempest-dsvm-intel-nfv-xenial/7a785f1 : SUCCESS in 40m 39s\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/35/402535/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/292e41a : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c1d049c07e002a2983eb6f957f385146224ab606","author":{"_account_id":16237,"name":"venkatamahesh","email":"venkatamaheshkotha@gmail.com","username":"venkatamahesh"},"date":"2016-11-27 02:07:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cc71e587f139f6d35586283d755b27d5e3057ad6","author":{"_account_id":20835,"name":"Gerald Lonlas","email":"glonlas@internap.com","username":"glonlas"},"date":"2016-12-05 22:43:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57a1b73ede377b8198c35595cb1e82b459cfb431","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2016-12-22 18:04:15.000000000","message":"Patch Set 1: Code-Review+2\n\nOooh, pretty badges","accounts_in_message":[],"_revision_number":1},{"id":"71c6bf0928d11acc959a2d801233e9cfad6429e1","author":{"_account_id":17922,"name":"Huan Xie","email":"huan.xie@citrix.com","username":"huan"},"date":"2016-12-27 10:12:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf2f4b9d1435d2b4d21dbf2fae9813e7c535f063","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2016-12-28 09:44:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"670ab304e3dd82fea3643aab6a43071d704d6afc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-28 09:44:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"c216fb1342396444a68dd7cec060d83a95de98db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-28 12:33:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/35/402535/1/gate/gate-nova-docs-ubuntu-xenial/7e12edc//doc/build/html/ : SUCCESS in 3m 13s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/35/402535/1/gate/gate-nova-pep8-ubuntu-xenial/549029c/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6b0abe806655522c896909a9fcf0b291b317375d","date":"2016-12-28 12:33:33.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e0cad906ae4209f6b1b017bce381b1c1a498245","revisions":{"3e0cad906ae4209f6b1b017bce381b1c1a498245":{"kind":"REWORK","_number":1,"created":"2016-11-25 12:58:26.000000000","uploader":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"ref":"refs/changes/35/402535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/402535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/402535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/402535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/402535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/402535/1"}}},"commit":{"parents":[{"commit":"2732d3e0f69c58697e464cb2bbb9bb8bffd3edd5","subject":"Merge \"Remove deprecated nova-all binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2732d3e0f69c58697e464cb2bbb9bb8bffd3edd5"}]}],"author":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-11-24 12:57:48.000000000","tz":60},"committer":{"name":"Flavio Percoco","email":"flaper87@gmail.com","date":"2016-11-25 12:55:29.000000000","tz":60},"subject":"Show team and repo badges on README","message":"Show team and repo badges on README\n\nThis patch adds the team\u0027s and repository\u0027s badges to the README file.\nThe motivation behind this is to communicate the project status and\nfeatures at first glance.\n\nFor more information about this effort, please read this email thread:\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-October/105562.html\n\nTo see an example of how this would look like check:\n\nb\u0027https://gist.github.com/596c8169411dd2535fd181ba4e2cb8ca\\n\u0027\n\nChange-Id: I2b02fc2968993947bec7e3b0e878b0ea149a9b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e0cad906ae4209f6b1b017bce381b1c1a498245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e0cad906ae4209f6b1b017bce381b1c1a498245"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2732d3e0f69c58697e464cb2bbb9bb8bffd3edd5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
