)]}'
{"id":"openstack%2Fnova~442105","triplet_id":"openstack%2Fnova~master~Iccd25339bc805b0b21a9290093338a010c204f4f","project":"openstack/nova","branch":"master","topic":"bug/1668432","hashtags":[],"change_id":"Iccd25339bc805b0b21a9290093338a010c204f4f","subject":"compute: Detach volumes on _rebuild_default_impl failure","status":"ABANDONED","created":"2017-03-06 18:41:28.000000000","updated":"2020-08-20 15:30:28.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b5421f24c97bf38118b53786aeb5ec7aa6adf55","_number":442105,"virtual_id_number":442105,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-01-09 17:41:06.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"date":"2018-01-09 18:29:50.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"date":"2020-07-31 22:01:30.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2018-01-09 17:45:18.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2020-07-27 14:44:00.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-09 17:41:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"date":"2020-08-05 01:05:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-07-27 12:07:34.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-27 13:50:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2018-01-10 10:09:11.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2018-01-09 20:07:22.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2018-01-09 18:57:50.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2020-07-27 11:48:22.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2018-01-09 19:36:51.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-30 18:46:48.000000000","updated_by":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-25 21:00:00.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-04-25 22:27:02.000000000","updated_by":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-04-26 03:08:43.000000000","updated_by":{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},"reviewer":{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},"state":"REVIEWER"},{"updated":"2017-08-10 13:19:59.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-08-10 15:03:41.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 17:41:06.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 17:41:29.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 17:45:18.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 18:29:50.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 18:57:50.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 19:36:51.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-09 20:07:22.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-10 10:09:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 11:48:22.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-27 12:07:34.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-07-27 13:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 14:44:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-31 22:01:30.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 01:05:48.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"}],"messages":[{"id":"0fe0dbe37cea6a11f29b0d7b95e9bf7bc4633bbe","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-03-06 18:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e4667658a9e997ff45575572d58b470861243a0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-03-06 19:16:00.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/442105/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e90882c922a024dd86ae18a94c1b2251d55d74e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-03-06 19:38:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/1/check/tempest-dsvm-full-xenial/520787a/ : SUCCESS in 56m 36s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/1/check/grenade-dsvm-xenial/b31f5aa/ : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b78b199a399ad6d0455b317ccf50574ac6145743","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-03-06 19:40:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/1/check/nova-out-of-tree-pvm/5bec1c8 : SUCCESS in 57m 40s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/1/check/nova-in-tree-pvm/998dae2 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3c28ecdf3d587c90b7e43a4d4c32f15fcbc20cd5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-03-06 19:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/442105/1/check/check-dsvm-tempest-vz7-exe-minimal/ae266e6 : SUCCESS in 59m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"942f34ad1eb97a9557dcd15504543e2e0aa64e44","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-03-06 19:58:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/442105/1/ : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"26806513ab77f02759621a6620ddc6adf670cad9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-03-06 20:00:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-06/442105/1/check/tempest-dsvm-full-nfv-xenial/795f612 : SUCCESS in 1h 18m 18s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-06/442105/1/check/tempest-dsvm-intel-nfv-xenial/bd874c6 : FAILURE in 36m 33s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-03-06/442105/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/2e7dd72 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":1},{"id":"334a1c9aef1f11804e6c03a4e77b6a630161d862","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-06 20:04:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/1/check/gate-nova-docs-ubuntu-xenial/b08aea9//doc/build/html/ : SUCCESS in 3m 00s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-nova-pep8-ubuntu-xenial/1f82b57/ : SUCCESS in 8m 48s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-nova-python27-ubuntu-xenial/187c445/ : SUCCESS in 11m 35s\n- gate-nova-python35 http://logs.openstack.org/05/442105/1/check/gate-nova-python35/1b900dd/ : SUCCESS in 11m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ce089fe/ : SUCCESS in 1h 14m 13s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7d1ae85/ : SUCCESS in 55m 39s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/95194eb/ : SUCCESS in 1h 12m 14s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-nova-tox-functional-ubuntu-xenial/ce532ef/ : SUCCESS in 15m 11s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/13f2fb7/ : SUCCESS in 14m 40s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8c6898d/ : SUCCESS in 41m 35s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/7b31792/ : SUCCESS in 41m 54s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/0af66e4/ : SUCCESS in 44m 04s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4d18f9a/ : FAILURE in 1h 12m 47s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/554362e/ : SUCCESS in 1h 17m 37s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/cd17804/ : SUCCESS in 1h 20m 06s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/52a2033/ : SUCCESS in 26m 20s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/4342058/ : SUCCESS in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ee5321d326465febaf49e743fa78f200ae526ff","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-03-06 20:21:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/1/check/dsvm-tempest-neutron-network/baf2217 : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"703c49919f9c1e73a1cfecfbb0eb1e9425e4e858","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2017-03-06 20:35:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-06/442105/1/check/tempest-dsvm-full-nfv-xenial/15b3625 : SUCCESS in 1h 53m 54s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-06/442105/1/check/tempest-dsvm-intel-nfv-xenial/35bd545 : FAILURE in 49m 32s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-06/442105/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/b61b18a : SUCCESS in 1h 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42006e102673fa21077a153b4c40e4f81e923e92","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-03-06 23:00:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/1/check/dsvm-tempest-kvm/0114445 : SUCCESS in 55m 33s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/1/check/dsvm-tempest-exe/6987e1d : SUCCESS in 44m 47s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"6683516d9865dc172ccb3a3f5b03c3a864f32d61","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-03-07 20:03:29.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/442105/1/check-ibm-tempest-dsvm-full/418b19c : FAILURE in 1h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"51e7c39b50e12123405810ebe7ddeba08cab1c0c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-03-21 10:59:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"fd62a4f37bb745ed452c29782ed6b624c9b80e78","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-03-21 11:52:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/2/check/tempest-dsvm-full-xenial/33ec1b3/ : SUCCESS in 52m 38s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/2/check/grenade-dsvm-xenial/cc05c65/ : SUCCESS in 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"431b09565768b9d3f154cec65e644870bb3f14b4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-03-21 12:23:13.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/442105/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4a7f98d827d31a6a4ee3d0bd26fc237d309743b7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-03-21 12:23:15.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/2/check/nova-out-of-tree-pvm/2f7abcc : FAILURE in 1h 23m 34s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/2/check/nova-in-tree-pvm/9ad157b : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":2},{"id":"84b34b21d0db3514db2ae876a339c15fc073d583","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-21 12:27:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/2/check/gate-nova-docs-ubuntu-xenial/c8b6908//doc/build/html/ : SUCCESS in 4m 34s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-nova-pep8-ubuntu-xenial/6fe58d1/ : SUCCESS in 8m 28s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-nova-python27-ubuntu-xenial/1fd10a8/ : SUCCESS in 10m 57s\n- gate-nova-python35 http://logs.openstack.org/05/442105/2/check/gate-nova-python35/e928a00/ : SUCCESS in 11m 33s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/afc14a4/ : SUCCESS in 1h 06m 01s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3bb79f4/ : SUCCESS in 48m 02s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/0a0680a/ : SUCCESS in 1h 14m 31s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-nova-tox-functional-ubuntu-xenial/5c2f0df/ : SUCCESS in 16m 32s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/d738aab/ : SUCCESS in 14m 21s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/025c424/ : SUCCESS in 53m 35s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/c687818/ : SUCCESS in 53m 49s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/940e621/ : SUCCESS in 50m 14s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b11bbf5/ : SUCCESS in 1h 10m 51s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e1aa609/ : SUCCESS in 1h 19m 00s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/2e1072d/ : SUCCESS in 1h 24m 41s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/fb4d4c5/ : SUCCESS in 25m 49s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/1c05eb0/ : SUCCESS in 43m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"692aedd835d779c1301df914ed8e56000ed2f43d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-03-21 12:39:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/2/check/dsvm-tempest-neutron-network/943f233 : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1f1bcac2aca89015716530cd0fc44725009d53a3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-03-21 13:17:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/442105/2/check/check-dsvm-tempest-vz7-exe-minimal/f0f1cc3 : SUCCESS in 1h 00m 04s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"9d644d3bc46d24c7c4aff235cc8ede8adfe3cbe7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-03-21 13:39:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-21/442105/2/check/tempest-dsvm-full-nfv-xenial/3f29822 : SUCCESS in 1h 32m 02s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-21/442105/2/check/tempest-dsvm-intel-nfv-xenial/19a7c31 : SUCCESS in 37m 38s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-03-21/442105/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/b34c954 : SUCCESS in 1h 09m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9e629ed794a1c2482da85e46131ccbc4efe1edc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-03-21 14:36:37.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-442105-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"f256ef6e11bf1bb8c50700ef0fabd8a187068a1f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-03-21 14:55:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/05/442105/2/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/dd67812 : SUCCESS in 1h 08m 10s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"e785819a44d6a831a7a8eef9cec9d6d82f780a74","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-03-21 15:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/2/check/dsvm-tempest-kvm/b6f4ed9 : SUCCESS in 1h 09m 32s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/2/check/dsvm-tempest-exe/870c630 : SUCCESS in 44m 47s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"220c52925ab132bb31cb4479e1c7589e0ceed7fd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-03-21 18:19:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/05/442105/2/check-nova/Nova-ML2-Sriov/8a9c992 : SUCCESS in 53m 39s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/05/442105/2/check-nova/Nova-MACVTAP-ML2-Sriov/7295af6 : SUCCESS in 41m 01s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"227e999ac2d59c93292230d7a55e632c3833e0ec","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-03-21 23:26:08.000000000","message":"Patch Set 2:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/442105/2/check-ibm-tempest-dsvm-full/67e75cd : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c75453cdaad3a3491ce59b2d4cb3da9b438f8298","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-03-30 15:06:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9e04efa62ea939134d47bb531c09b254a84753b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-03-30 15:57:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/3/check/tempest-dsvm-full-xenial/531e4f8/ : SUCCESS in 50m 10s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/3/check/grenade-dsvm-xenial/ad7fc06/ : SUCCESS in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb8c91db0da3a36ebb0d3b509a2ea604cfa4b416","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-03-30 16:03:26.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/442105/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2c6b9b8f968db4b72f4b280bf40dc9bfd43b67aa","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-03-30 16:17:03.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/442105/3/ : FAILURE in 1h 09m 31s","accounts_in_message":[],"_revision_number":3},{"id":"81c46990c911e6a1686c6244d90622bd76b98a41","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-03-30 16:32:33.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/3/check/nova-out-of-tree-pvm/82e9fca : FAILURE in 1h 25m 03s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/3/check/nova-in-tree-pvm/87479f0 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":3},{"id":"416993e9e27bf30c865a2870b86c7244797921d9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-30 16:35:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/3/check/gate-nova-docs-ubuntu-xenial/2ae93b5//doc/build/html/ : SUCCESS in 5m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-nova-pep8-ubuntu-xenial/e85f751/ : SUCCESS in 10m 38s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-nova-python27-ubuntu-xenial/25823e3/ : SUCCESS in 12m 31s\n- gate-nova-python35 http://logs.openstack.org/05/442105/3/check/gate-nova-python35/3f71388/ : SUCCESS in 11m 59s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/592e6bc/ : SUCCESS in 1h 16m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a7a0641/ : SUCCESS in 51m 59s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/e1ee72c/ : SUCCESS in 57m 56s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-nova-tox-functional-ubuntu-xenial/be64b72/ : SUCCESS in 15m 32s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-nova-tox-functional-py35-ubuntu-xenial/0bb3495/ : SUCCESS in 16m 04s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/aa3764a/ : SUCCESS in 39m 46s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/3/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/3896dd0/ : SUCCESS in 53m 01s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/94a49b9/ : SUCCESS in 44m 37s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9c953df/ : SUCCESS in 1h 06m 56s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/fae333c/ : SUCCESS in 1h 12m 55s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/7dcbe8f/ : SUCCESS in 1h 26m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0552802/ : SUCCESS in 25m 26s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/8537404/ : SUCCESS in 37m 58s (non-voting)\n- gate-placement-api-ref-nv http://logs.openstack.org/05/442105/3/check/gate-placement-api-ref-nv/af74efe/ : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90aa7bfab0b9815994c7716285b08a9d7d24e1f2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-03-30 16:43:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-30/442105/3/check/tempest-dsvm-full-nfv-xenial/676c352 : SUCCESS in 1h 36m 05s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-03-30/442105/3/check/tempest-dsvm-intel-nfv-xenial/c6cff0a : SUCCESS in 38m 53s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-03-30/442105/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/20bb9d3 : SUCCESS in 1h 21m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfbc11a340beae098b40314bb164ed111fc994c1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-03-30 16:43:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/442105/3/check/check-dsvm-tempest-vz7-exe-minimal/c5d3b59 : SUCCESS in 1h 24m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"92b069dbc192a2e675971673ff6dd4f4dd4f2725","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-03-30 16:49:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/3/check/dsvm-tempest-neutron-network/41e2c37 : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":3},{"id":"808bc3613d84e2f8fdb21cb061d45385e9abdfb0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-03-30 17:28:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/05/442105/3/check/dsvm-tempest-xen/71baeed : SUCCESS in 2h 10m 57s","accounts_in_message":[],"_revision_number":3},{"id":"68d87d79c7383342f26ea1dbb19501c6a368f07c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-03-30 18:32:32.000000000","message":"Patch Set 3:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/442105/3/check-ibm-tempest-dsvm-full/37ea2c4 : FAILURE in 2h 23m 08s","accounts_in_message":[],"_revision_number":3},{"id":"ff48813b6cb46aed73ddc4f459bf9e41b916f13f","author":{"_account_id":16897,"name":"Virtuozzo Storage CI","email":"vzstorage-ci@virtuozzo.com","username":"vzstorage-ci","tags":["SERVICE_USER"]},"date":"2017-03-30 18:46:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-kvm http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/3/check/dsvm-tempest-kvm/f4cf50e : SUCCESS in 57m 31s\n- dsvm-tempest-exe http://openstack-3rd-party-storage-ci-logs.virtuozzo.com/05/442105/3/check/dsvm-tempest-exe/1991ea1 : SUCCESS in 49m 39s (non-voting)\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_Storage_CI Make the comment \u0027run-Virtuozzo Storage CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"76281a1f459788c8762852e2b3468239dff34d32","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2017-03-30 19:54:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-30/442105/3/check/tempest-dsvm-full-nfv-xenial/dae2853 : SUCCESS in 1h 42m 38s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-30/442105/3/check/tempest-dsvm-intel-nfv-xenial/1b48310 : SUCCESS in 51m 09s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-03-30/442105/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e6ae251 : FAILURE in 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"148935c6394e9f7b8828eaf3d4e8dcebf3c6b30f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-03-30 21:30:34.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-442105-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"0db0ea3696bf011575b114c65e7880383eb65ae7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-03-30 23:04:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/05/442105/3/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/5856cc6 : SUCCESS in 1h 01m 00s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"9463e94d45ad31c62fa06498cd9e708a12638c21","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-03-31 09:30:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/05/442105/3/check-nova/Nova-ML2-Sriov/257ebac : SUCCESS in 53m 08s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/05/442105/3/check-nova/Nova-MACVTAP-ML2-Sriov/d3dbcdc : SUCCESS in 47m 30s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"0e2f6519d829dd7d30cb53fc366ce095c04dab39","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-04-21 09:42:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1a15f4853984ad245ba2c3bd6fb1b2b0eb30bd95","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-04-21 10:24:07.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/442105/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"92841a641895a928c0c955791aeabe063edb49bc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-04-21 10:30:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/4/check/tempest-dsvm-full-xenial/8f5fd49/ : SUCCESS in 47m 05s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/4/check/grenade-dsvm-xenial/857ca8c/ : SUCCESS in 38m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"169090a0386442cf6f52ce07e60b9ea6663403f0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-04-21 11:02:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-21/442105/4/check/tempest-dsvm-full-nfv-xenial/e3ef989 : SUCCESS in 1h 18m 38s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-21/442105/4/check/tempest-dsvm-intel-nfv-xenial/b935277 : SUCCESS in 35m 16s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-04-21/442105/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/8facd8e : SUCCESS in 1h 09m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"479929eac63eb0d9c5365dd81938f540ff4f5ac6","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-04-21 11:04:39.000000000","message":"Patch Set 4:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/442105/4/check-ibm-tempest-dsvm-full/8d68a0f : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":4},{"id":"cdb75b201798a8bdafc97203c9de27e7051fe7d3","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-04-21 11:06:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/442105/4/ : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":4},{"id":"cfb6ed9e8f8f5c867ca0a95143ff97d1ce67ceac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-21 11:17:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/4/check/gate-nova-docs-ubuntu-xenial/3d5bf4c//doc/build/html/ : SUCCESS in 4m 43s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-nova-pep8-ubuntu-xenial/67e1df4/ : SUCCESS in 6m 05s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-nova-python27-ubuntu-xenial/5c66aed/ : SUCCESS in 10m 46s\n- gate-nova-python35 http://logs.openstack.org/05/442105/4/check/gate-nova-python35/b0b26af/ : SUCCESS in 10m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/b4185d1/ : SUCCESS in 1h 15m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/6615b60/ : SUCCESS in 34m 24s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/d2dada7/ : SUCCESS in 1h 07m 00s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-nova-tox-functional-ubuntu-xenial/7638b42/ : SUCCESS in 13m 52s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-nova-tox-functional-py35-ubuntu-xenial/2830f8b/ : SUCCESS in 16m 16s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/659fc78/ : SUCCESS in 39m 12s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2bfe071/ : SUCCESS in 36m 48s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/18f2e5c/ : SUCCESS in 52m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/98c7685/ : SUCCESS in 1h 03m 07s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/9fdfe7e/ : SUCCESS in 1h 31m 02s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/80852fd/ : SUCCESS in 1h 06m 08s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5e8d723/ : SUCCESS in 22m 44s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/78698fc/ : SUCCESS in 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5036d763e1bab2ea08ed8e52a20b104b760a9387","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-04-21 11:21:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/4/check/nova-out-of-tree-pvm/3e048bf : SUCCESS in 1h 38m 14s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/4/check/nova-in-tree-pvm/96027a7 : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":4},{"id":"859970eb5d258b32a52052b9d5041c4667622792","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-04-21 11:24:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/4/check/dsvm-tempest-neutron-network/7dff143 : FAILURE in 1h 41m 22s","accounts_in_message":[],"_revision_number":4},{"id":"685643d84347616114879788aa3dfd3d38d6924a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-04-21 11:44:24.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/05/442105/4/check/dsvm-tempest-xen/5d900bc : FAILURE in 2h 01m 18s","accounts_in_message":[],"_revision_number":4},{"id":"7f5f89d83e186747a7f2074e8249656ddf2cfe28","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-04-21 11:53:03.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-442105-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"da9a4101409ae42e034ccbc7a6a7220dd3162907","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-04-21 12:13:42.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/05/442105/4/check/ext-nova-zuul/55d7741 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":4},{"id":"307aeb03236b4d5a90fb72d45cdd1390d01a15f6","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-04-25 19:25:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7fcc5bbb4471cb51d69558f5d2b7db90382b008a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-04-25 20:12:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/5/check/tempest-dsvm-full-xenial/50b0c97/ : SUCCESS in 46m 38s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/5/check/grenade-dsvm-xenial/3e54a39/ : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3cd158bb0fd39b193c12937e6e3db8d90cb950a3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-04-25 20:24:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/5/check/nova-out-of-tree-pvm/1118d3c : SUCCESS in 58m 10s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/5/check/nova-in-tree-pvm/a5896a0 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":5},{"id":"9e79d4225b5e002f7f2c35b4af2bf34962729a10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-25 20:54:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/5/check/gate-nova-docs-ubuntu-xenial/8dac5aa//doc/build/html/ : SUCCESS in 3m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-nova-pep8-ubuntu-xenial/d633f18/ : SUCCESS in 6m 05s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-nova-python27-ubuntu-xenial/1f73a38/ : SUCCESS in 9m 53s\n- gate-nova-python35 http://logs.openstack.org/05/442105/5/check/gate-nova-python35/311ecae/ : SUCCESS in 11m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/49658a3/ : SUCCESS in 1h 08m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-grenade-dsvm-neutron-ubuntu-xenial/6ca7a92/ : SUCCESS in 37m 00s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-py35-ubuntu-xenial/661abbb/ : SUCCESS in 1h 05m 46s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-nova-tox-functional-ubuntu-xenial/d175984/ : SUCCESS in 28m 35s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-nova-tox-functional-py35-ubuntu-xenial/f7152f4/ : SUCCESS in 27m 22s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/2755279/ : SUCCESS in 42m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/5/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/dddc543/ : SUCCESS in 39m 31s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-cells-ubuntu-xenial/cf6fce9/ : SUCCESS in 38m 49s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/56fd916/ : FAILURE in 1h 00m 23s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3600a45/ : SUCCESS in 1h 26m 31s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/c83bb4a/ : SUCCESS in 1h 06m 57s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/4b9f3f9/ : SUCCESS in 25m 19s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/5/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/baef021/ : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d44300ca9f5aacf421982443d6809762dc3f1326","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-04-25 21:00:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/442105/5/ : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2daedea36be31769963893e04d77e65dd92e94d5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-04-25 21:02:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/5/check/dsvm-tempest-neutron-network/3547089 : SUCCESS in 1h 36m 20s","accounts_in_message":[],"_revision_number":5},{"id":"3cec7c69531a9f7057159c938a4a4428f4fb99fa","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-04-25 21:16:17.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/442105/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"c5977d5d40d86e33fd3108b77d9aef6a9b08a404","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-04-25 22:18:40.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/05/442105/5/check/dsvm-tempest-xen/f1b1c11 : FAILURE in 2h 14m 52s","accounts_in_message":[],"_revision_number":5},{"id":"da94f8d05de6d8ee545b28ad388184b1187fb11e","author":{"_account_id":14571,"name":"Intel Experimental CI","email":"openstack-networking-ci@intel.com","username":"intel-networking-ci","tags":["SERVICE_USER"]},"date":"2017-04-25 22:27:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-25/442105/5/check/tempest-dsvm-full-nfv-xenial/1bd92b7 : SUCCESS in 1h 44m 49s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-25/442105/5/check/tempest-dsvm-intel-nfv-xenial/c85ac60 : SUCCESS in 46m 47s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-04-25/442105/5/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/f96506d : SUCCESS in 1h 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"53d3d969787cb2ffdcbd95141df7054721eacb5a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-04-25 22:34:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-25/442105/5/check/tempest-dsvm-full-nfv-xenial/4af865e : SUCCESS in 1h 22m 04s (non-voting)\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/2017-04-25/442105/5/check/tempest-dsvm-intel-nfv-xenial/6931c47 : SUCCESS in 33m 49s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/2017-04-25/442105/5/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/dd0fb6d : SUCCESS in 1h 06m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f6410dcc5c781b2bd38f6972bcd93652d29cf3ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-04-26 02:03:42.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/05/442105/5/check/ext-nova-zuul/c59b71d : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":5},{"id":"e1ea20aa2a3f3e48aa500d27939d205924a720a3","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-04-26 02:56:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/05/442105/5/check-nova/Nova-ML2-Sriov/49385a9 : SUCCESS in 45m 47s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/05/442105/5/check-nova/Nova-MACVTAP-ML2-Sriov/d3fc0d2 : SUCCESS in 52m 52s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"5b1a08aaf9100a54f4e99e55a079c9efbec0ef3b","author":{"_account_id":17814,"name":"fengzhr","email":"fengzhr@awcloud.com","username":"space"},"date":"2017-04-26 03:08:43.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3c4927364f61e035290f5552520b3b9cc2ee3b42","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-04-26 03:13:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://13.69.151.247/05/442105/5/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/41f7226 : SUCCESS in 1h 04m 02s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"6c9e05eace24b6b3e0213b2d816d1a8cb3b6e224","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-04-26 03:32:55.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-442105-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"bb7a409123cb26ec72fc2ac5abf4be9569776d48","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-08-10 12:54:25.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"20c291766865731f95a66b2b62a1ed961db26ce2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-10 13:19:59.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/442105/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"0d39f6bf5a40387a579138d3c3037e2dc2baacd0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-10 14:00:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/6/check/tempest-dsvm-full-xenial/6ad0339/ : SUCCESS in 1h 05m 29s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/6/check/grenade-dsvm-xenial/14f8c58/ : SUCCESS in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c4382cd5b181f0674b47e017364b3884f20ff6f5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-10 14:20:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/05/442105/6/check-nova/Nova-ML2-Sriov/8d71205 : SUCCESS in 1h 01m 34s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/05/442105/6/check-nova/Nova-MACVTAP-ML2-Sriov/219a761 : SUCCESS in 53m 13s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"96cbf4d7927c9ac4a7018310680e107e50128707","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-10 14:31:25.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/6/check/nova-out-of-tree-pvm/e16cc14 : SUCCESS in 1h 36m 06s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/6/check/nova-in-tree-pvm/d50fc34 : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":6},{"id":"73f7a6295eabe21bb054e1ff8e66056e8ada8e0c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-10 14:57:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/442105/6/ : SUCCESS in 1h 49m 13s","accounts_in_message":[],"_revision_number":6},{"id":"36db362f06f5435c2debc68853f70d5ba87f00ed","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-10 15:03:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-10/442105/6/check/tempest-dsvm-full-nfv-xenial/c2e0f5b : SUCCESS in 2h 08m 43s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-10/442105/6/check/tempest-dsvm-intel-nfv-xenial/f2c4cf6 : SUCCESS in 49m 17s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-10/442105/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/dff1796 : SUCCESS in 1h 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed3e8c8265f2e68a56f942cf0f6583d7b69ca74a","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-10 15:24:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/05/442105/6/check/dsvm-tempest-xen/2a88f27 : SUCCESS in 2h 22m 40s","accounts_in_message":[],"_revision_number":6},{"id":"61e4b249a20183b2623586f781bcfc7a95d64cec","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-10 15:44:51.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/442105/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/8fd46cf : SUCCESS in 2h 48m 12s","accounts_in_message":[],"_revision_number":6},{"id":"ef4867c5b709a35ac51f4e55353844e43ffe44ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-10 20:02:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/05/442105/6/check/gate-nova-docs-ubuntu-xenial/56785f3//doc/build/html/ : SUCCESS in 5m 14s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-nova-pep8-ubuntu-xenial/0862227/ : SUCCESS in 8m 24s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-nova-python27-ubuntu-xenial/042ad0b/ : SUCCESS in 16m 20s\n- gate-nova-python35 http://logs.openstack.org/05/442105/6/check/gate-nova-python35/71a291a/ : SUCCESS in 11m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/176ff3f/ : FAILURE in 3h 09m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ce88772/ : SUCCESS in 46m 31s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-py35-ubuntu-xenial/521337d/ : SUCCESS in 1h 33m 19s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-nova-tox-functional-ubuntu-xenial/b5621fb/ : SUCCESS in 23m 50s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-nova-tox-functional-py35-ubuntu-xenial/85d7baa/ : SUCCESS in 30m 45s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/eb3bcc7/ : SUCCESS in 58m 21s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/05/442105/6/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2dee440/ : SUCCESS in 1h 06m 33s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-cells-ubuntu-xenial/371e121/ : SUCCESS in 1h 18m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a39b79d/ : SUCCESS in 1h 16m 07s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/71840b5/ : FAILURE in 1h 20m 25s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/5e61ace/ : SUCCESS in 1h 17m 20s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/98b39bc/ : SUCCESS in 31m 31s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/05/442105/6/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/13655c0/ : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ee71e963f890b8e961ae2732d4c1add04ab7b20a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-08-21 11:43:38.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : EXCEPTION\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":6},{"id":"9eecd1ff39af36d3d6952fca8a914e8b3093cbdb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 02:28:21.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":6},{"id":"159b8f6310ba3872797ebc0414731a5ca36f5fce","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2018-01-09 16:43:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8156691d242f1fa8557ad87cfa61bcf112a3f38e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-01-09 17:37:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/05/442105/7/check-nova/Nova-ML2-Sriov/5e3f946 : SUCCESS in 49m 34s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/05/442105/7/check-nova/Nova-MACVTAP-ML2-Sriov/2bc2b80 : SUCCESS in 47m 18s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"266f3e216809e3dad3c6ca2ac14fc8128ac539f9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-09 17:41:06.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/05/442105/7/check/nova-out-of-tree-pvm/f348ecd : SUCCESS in 51m 53s\n- nova-in-tree-pvm http://184.172.12.213/05/442105/7/check/nova-in-tree-pvm/c29c143 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":7},{"id":"a32862b978a2dbf31f9ae76d27dc3a3f396abf12","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-09 17:41:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/7/check/tempest-dsvm-full-xenial/cb1d5c2/ : SUCCESS in 56m 45s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/05/442105/7/check/grenade-dsvm-xenial/ff4689a/ : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5980418b91452525059e3c9b2f90d3a09bf4c1f4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-01-09 17:45:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://52.169.200.208/05/442105/7/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/92145d0 : SUCCESS in 59m 09s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"907c3496ef4195f47aa0ea0084f3bab23a8cfd1b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-01-09 17:55:13.000000000","message":"Patch Set 7:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/442105/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f7d648e : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":7},{"id":"7e582956a22d7dd3f4ef3fd0abbfc92163898a89","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-09 18:29:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/05/442105/7/check/check-dsvm-tempest-vz7-exe-minimal/5b871a6 : SUCCESS in 52m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"81e30a0266268703d08450d04179691a730e94e1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-01-09 18:57:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/7/check/dsvm-tempest-neutron-network/f4cdae6 : SUCCESS in 2h 12m 36s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/05/442105/7/check/test-vgpu/1db978e : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"91482fdb09999800b4060296bedc7b75d8c4eec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-09 19:06:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/05/442105/7/check/openstack-tox-pep8/45e82b0/ : SUCCESS in 11m 12s\n- openstack-tox-py27 http://logs.openstack.org/05/442105/7/check/openstack-tox-py27/99db12f/ : SUCCESS in 12m 51s\n- openstack-tox-py35 http://logs.openstack.org/05/442105/7/check/openstack-tox-py35/43d6d7a/ : SUCCESS in 22m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/442105/7/check/build-openstack-sphinx-docs/9c87181/html/ : SUCCESS in 6m 15s\n- tempest-full http://logs.openstack.org/05/442105/7/check/tempest-full/7f98a26/ : SUCCESS in 1h 43m 50s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/05/442105/7/check/legacy-grenade-dsvm-neutron/7c3b720/ : SUCCESS in 57m 08s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-py35/3809236/ : SUCCESS in 1h 14m 11s\n- openstack-tox-functional http://logs.openstack.org/05/442105/7/check/openstack-tox-functional/de2702d/ : SUCCESS in 24m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/05/442105/7/check/openstack-tox-functional-py35/e83fbc7/ : SUCCESS in 23m 17s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/05/442105/7/check/legacy-grenade-dsvm-neutron-multinode/a2fba5c/ : SUCCESS in 1h 01m 16s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/05/442105/7/check/legacy-grenade-dsvm-neutron-multinode-live-migration/d585f9f/ : SUCCESS in 56m 09s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-cells/fad7c54/ : SUCCESS in 36m 32s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/501dabe/ : SUCCESS in 1h 03m 07s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-neutron-multinode-full/cfecd83/ : SUCCESS in 1h 24m 40s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-neutron-nova-next-full/5ad0e52/ : SUCCESS in 1h 11m 43s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/05/442105/7/check/legacy-tempest-dsvm-multinode-live-migration/f45b42d/ : SUCCESS in 44m 26s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/05/442105/7/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/64e8a85/ : SUCCESS in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1b9757cae94bd52b43b85b978e4cb34f643b62b0","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2018-01-09 19:36:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/05/442105/7/check/dsvm-tempest-xen/e7beed8 : SUCCESS in 2h 52m 05s","accounts_in_message":[],"_revision_number":7},{"id":"37efe47943680880e64fd77b5d18996d70bb0112","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-01-09 20:07:22.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-05-442105-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"49740535fe5b0409cc6bc949fd1556e936be6038","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-09 20:25:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/442105/7/ : SUCCESS in 2h 34m 09s","accounts_in_message":[],"_revision_number":7},{"id":"467fcff41f2aadbb925ebed1772a30cde3e36419","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-10 10:09:11.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/05/442105/7/check-vote/ext-nova-zuul/58dab9b : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":7},{"id":"0edc1e77e181af32a146e2cf9870bfbf30d6d298","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2018-07-06 13:43:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7},{"id":"ab71d2b2edf8886cc6d9861311134b9e81041cb9","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-27 11:18:01.000000000","message":"Restored","accounts_in_message":[],"_revision_number":7},{"id":"7f7de77439ec4a4ee19648cf72d5bd00e54ce494","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-27 11:18:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/05/442105/7/check/EMC_PowerFlex_NOVA/cb00b80/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"d6383d138ce181762638a8750d49b3aaea8e523b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-27 11:48:22.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/442105/7 : FAILURE in 27m 28s","accounts_in_message":[],"_revision_number":7},{"id":"6a83f2cd91f67ea84a3e5d6bd981d45edccffe58","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-27 12:07:34.000000000","message":"Patch Set 7:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/05/442105/7/check/pmem-tempest-plugin-filtered-multi-node/dbbe13b/ : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":7},{"id":"259725169924ee56217300db922079f34d1fff19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 13:50:46.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/042b0401378e47c4a444b2569f9bd9e9 : SUCCESS in 1h 11m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7327491cd78b4a8bbf1b49f29fa0c054 : SUCCESS in 1h 30m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ccf1ffd2ef0436daf934d2ad7bf8c72 : FAILURE in 16m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15abb3d2eb594f2689e709400e97298a : FAILURE in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52922f866b24a678bf63580130b9275 : FAILURE in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99929c554c0a41a49dc5d8ac3fd4af44 : FAILURE in 14m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69cbd9c6427f412fb412980460fb0841 : FAILURE in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03bce5d612754ee9ab5d54759700098b : SUCCESS in 11m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/581bf2303114463b95f3382c93f6eba2 : SUCCESS in 56m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a82f3eea04d44e1389c7cec744ffb5f3 : SUCCESS in 1h 14m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab606f96bafa4de793b188c82e83a5ea : SUCCESS in 1h 25m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b2f046df38b649189e701ff72080e057 : SUCCESS in 56m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c545385025c24fbeb3a3876553be492b : SUCCESS in 1h 16m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/fb6fbd23cd0d44a8b1b4134f823861a4 : SUCCESS in 1h 37m 18s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a31e16d629b4b939780d95eee0dea43 : SUCCESS in 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e46f9e167424cf3a376965dbdad6b6a : FAILURE in 1h 00m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a43fb7e9ebfd40c3a731d827c6ff0a91 : SUCCESS in 1h 02m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1d2bbc96d2bc4ab08d474c6337504749 : SUCCESS in 30m 03s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/5860a20837514fd98c2ee06f51c1d67d : SUCCESS in 42m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"769bc22be94f6221881e831723b556d7d2771e87","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-27 14:43:31.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/442105/7/check/tempest-dsvm-full-bionic/75478d3/ : SUCCESS in 1h 38m 52s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/442105/7/check/tempest-dsvm-full-bionic-py3/73978c2/ : SUCCESS in 1h 36m 10s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/442105/7/check/grenade-dsvm-bionic/39a9a75/ : SUCCESS in 2h 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"45b1db98fab0c475f699a5b056bc6b00efd95a50","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-07-27 14:44:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/05/442105/7/check/EMC_PowerFlex_NOVA/5913e6e/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"0bf54c05418585ef8169486dc37cac3e4a6ffc92","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-07-31 22:01:30.000000000","message":"Patch Set 7:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/442105/7/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/162ccd7 : SUCCESS in 1h 36m 02s","accounts_in_message":[],"_revision_number":7},{"id":"aa726232bd23d9316ead8ee2f85a29a32c27588b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-08-05 01:05:48.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-44086 : FAILURE in 1h 16m 19s","accounts_in_message":[],"_revision_number":7},{"id":"8219f3f1b09221ca39426426a6c7c4d198b5ad34","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-20 15:30:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4ab416addec436870355421c5d02a20633df7fba","revisions":{"071f75fa070c04c065ce670d8cb910dc1f5d7d0e":{"kind":"REWORK","_number":1,"created":"2017-03-06 18:41:28.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/1"}}},"commit":{"parents":[{"commit":"5cf6bbf374a8b877d2e158aa8802b31d14ceb121","subject":"Merge \"Add warning on setting secure_proxy_ssl_header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cf6bbf374a8b877d2e158aa8802b31d14ceb121"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 18:36:09.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/071f75fa070c04c065ce670d8cb910dc1f5d7d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/071f75fa070c04c065ce670d8cb910dc1f5d7d0e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5cf6bbf374a8b877d2e158aa8802b31d14ceb121","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"659ba534dd61435932e9d5e47d1badefcdd1629d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-03-21 10:59:03.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/2"}}},"commit":{"parents":[{"commit":"1e0e4080fbe0515e6c0734ae3907d6a5d0e3079f","subject":"Merge \"DELETE all inventory for a resource provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e0e4080fbe0515e6c0734ae3907d6a5d0e3079f"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-21 10:59:02.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/659ba534dd61435932e9d5e47d1badefcdd1629d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/659ba534dd61435932e9d5e47d1badefcdd1629d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1e0e4080fbe0515e6c0734ae3907d6a5d0e3079f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"07ad5fd6a60725f1f70196d46281cfff13f79e10":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-03-30 15:06:46.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/3"}}},"commit":{"parents":[{"commit":"837b039b0967e54b4a3374cdaabf9df649b809cd","subject":"Merge \"numa: update numa usage to include reserved CPUs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/837b039b0967e54b4a3374cdaabf9df649b809cd"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-30 15:06:45.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07ad5fd6a60725f1f70196d46281cfff13f79e10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07ad5fd6a60725f1f70196d46281cfff13f79e10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"837b039b0967e54b4a3374cdaabf9df649b809cd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"43da026218582134da18682d667226473108195d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-04-21 09:42:31.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/4"}}},"commit":{"parents":[{"commit":"3cc34c2aa3d7bae8863dd3238ecea2765bb6c855","subject":"Merge \"PowerVM Driver: spawn/destroy #2: functional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cc34c2aa3d7bae8863dd3238ecea2765bb6c855"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-04-21 09:42:31.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43da026218582134da18682d667226473108195d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43da026218582134da18682d667226473108195d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3cc34c2aa3d7bae8863dd3238ecea2765bb6c855","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"cceda0389472aca390d0320dae25d2abbbe1aca6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2017-04-25 19:25:20.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/5"}}},"commit":{"parents":[{"commit":"ec1794b18c6ecb9b0c1e6a5f3908db70f4531328","subject":"Merge \"api-ref: move createBackup to server-actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec1794b18c6ecb9b0c1e6a5f3908db70f4531328"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-04-25 19:25:20.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cceda0389472aca390d0320dae25d2abbbe1aca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cceda0389472aca390d0320dae25d2abbbe1aca6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ec1794b18c6ecb9b0c1e6a5f3908db70f4531328","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3501ac6c8a540a358742b1b2297f182167fea79c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-08-10 12:54:25.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/6"}}},"commit":{"parents":[{"commit":"f342a42550326b81ad2a625f1aa7e1dd41895714","subject":"Merge \"Add policy granularity to the Flavors API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f342a42550326b81ad2a625f1aa7e1dd41895714"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-08-10 12:54:25.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3501ac6c8a540a358742b1b2297f182167fea79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3501ac6c8a540a358742b1b2297f182167fea79c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f342a42550326b81ad2a625f1aa7e1dd41895714","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4ab416addec436870355421c5d02a20633df7fba":{"kind":"REWORK","_number":7,"created":"2018-01-09 16:43:57.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/442105/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/442105/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/442105/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/442105/7"}}},"commit":{"parents":[{"commit":"289d623d8e6598d988f1341e3dc8627adc2e043a","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/289d623d8e6598d988f1341e3dc8627adc2e043a"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2017-03-06 17:02:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2018-01-09 16:43:14.000000000","tz":0},"subject":"compute: Detach volumes on _rebuild_default_impl failure","message":"compute: Detach volumes on _rebuild_default_impl failure\n\nPreviously a failed call to spawn an instance during a rebuild using the\ndefault implementation would leave volumes accessible from and\npotentially connected to compute hosts. This change now ensures the\nprovided detach_block_devices method is called again if spawn fails.\n\nCloses-bug: #1668432\nChange-Id: Iccd25339bc805b0b21a9290093338a010c204f4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ab416addec436870355421c5d02a20633df7fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ab416addec436870355421c5d02a20633df7fba"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"289d623d8e6598d988f1341e3dc8627adc2e043a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
