)]}'
{"id":"openstack%2Fnova~461448","triplet_id":"openstack%2Fnova~master~I21f4b18e4554e9e912496e623461129196c72852","project":"openstack/nova","branch":"master","topic":"bp/service-hyper-uuid-in-ap","hashtags":[],"change_id":"I21f4b18e4554e9e912496e623461129196c72852","subject":"objects: add ComputeNode.get_by_uuid method","status":"MERGED","created":"2017-05-01 16:37:14.000000000","updated":"2017-05-15 10:38:43.000000000","submitted":"2017-05-15 10:38:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d01bcd08b26c0777b7a6cab48491fef7c8a933c","_number":461448,"virtual_id_number":461448,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-05-15 10:38:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2017-05-04 18:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2017-05-04 20:21:25.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"recommended":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":1,"date":"2017-05-10 22:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2017-05-11 05:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2017-05-15 09:22:20.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2017-05-05 14:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2017-05-15 09:22:20.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-01 16:58:35.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"},{"updated":"2017-05-01 18:00:51.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-01 18:15:18.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-02 03:22:41.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-05-02 17:35:22.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-02 19:05:48.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-02 19:15:34.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-02 20:11:52.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-03 00:48:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-03 02:38:51.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-03 02:52:05.000000000","updated_by":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"reviewer":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"state":"REVIEWER"},{"updated":"2017-05-04 17:20:48.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-05-04 17:31:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-04 18:52:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-04 20:21:25.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-05-05 14:45:58.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2017-05-10 22:33:57.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2017-05-11 05:07:58.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-05-15 09:22:20.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2017-05-15 10:38:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f032047f04e86eb1f3f34cf7fd1e4c1881455c8c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-01 16:37:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e2ca548ff731c47da42b66cca02568d4f009be0","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2017-05-01 16:58:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f7c97fce3d9c92e39d18eee0bbb8b5498a9b591","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-01 17:28:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/tempest-dsvm-full-xenial/4c313b4/ : SUCCESS in 48m 43s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/grenade-dsvm-xenial/ea1ae09/ : SUCCESS in 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c01d643d349adcdcd498b0e755dc4e7a8bea0b22","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-05-01 17:40:42.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/461448/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"21152bd70caeea4d6675e7016a29e664d84c5044","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-01 17:48:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/1/ : FAILURE in 1h 09m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a0d1601bb13a3ee0374d4aa21e852dc2613e5ff7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-05-01 18:00:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/48/461448/1/check/nova-out-of-tree-pvm/f539bda : SUCCESS in 1h 14m 00s\n- nova-in-tree-pvm http://184.172.12.213/48/461448/1/check/nova-in-tree-pvm/7987265 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a825e88e1fa3a31394f579d6102ca3f1a9f8ca1d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-05-01 18:15:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/48/461448/1/check/check-dsvm-tempest-vz7-exe-minimal/f7d0872 : SUCCESS in 1h 12m 45s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"fb82e8767497483470b8b3b5f6bea3979afb91d9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-01 18:32:13.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461448/1/check-ibm-tempest-dsvm-full/d55ab33 : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c5a6edb2280fe797aa6db02214b6dead09cbd2de","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-01 18:33:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/1/check/gate-nova-docs-ubuntu-xenial/24301ac//doc/build/html/ : SUCCESS in 3m 59s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-pep8-ubuntu-xenial/451676a/ : SUCCESS in 6m 08s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-python27-ubuntu-xenial/d29c9c4/ : SUCCESS in 12m 02s\n- gate-nova-python35 http://logs.openstack.org/48/461448/1/check/gate-nova-python35/b28eee8/ : SUCCESS in 10m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d407322/ : SUCCESS in 1h 04m 41s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/b49de5f/ : SUCCESS in 58m 17s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/8f0b352/ : FAILURE in 50m 05s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-ubuntu-xenial/2820484/ : SUCCESS in 31m 40s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/f17d61a/ : SUCCESS in 21m 53s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/64c847b/ : SUCCESS in 1h 00m 27s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a93fc04/ : SUCCESS in 57m 33s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/6cdb08e/ : SUCCESS in 50m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5e09021/ : FAILURE in 1h 13m 54s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e9a9adf/ : SUCCESS in 1h 48m 04s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/838919c/ : FAILURE in 51m 28s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5d52e13/ : SUCCESS in 41m 23s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/ea2800d/ : SUCCESS in 44m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8738e58f57f3df291f888ce6cbb4a05e22dd5381","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-05-01 18:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/48/461448/1/check/dsvm-tempest-neutron-network/4231fa4 : SUCCESS in 1h 46m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8c8390bbf6479dfcb420f828a6407de2f9a716c7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-05-01 18:59:04.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/48/461448/1/check/dsvm-tempest-xen/190aaf1 : FAILURE in 2h 08m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5c979326e6aa9bc767d5dfeb219d2e8c9d45ae1b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-05-01 19:17:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/48/461448/1/check-nova/Nova-ML2-Sriov/dba7dd2 : SUCCESS in 52m 58s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/48/461448/1/check-nova/Nova-MACVTAP-ML2-Sriov/d1fded1 : FAILURE in 51m 52s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"f5f3d14884dcbfa8594f0269958facef59291dce","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-01 21:45:12.000000000","message":"Patch Set 1:\n\nrecheck bug 1687044","accounts_in_message":[],"_revision_number":1},{"id":"ff3e098e95097b417952187c0f1d4a535705c07e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-01 21:52:35.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/tempest-dsvm-full-xenial/9bfddb6/ : FAILURE in 6m 17s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/grenade-dsvm-xenial/128a44b/ : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45e2a5328ebd5cc729e50bb75619aa739301a3d5","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-01 22:54:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/1/ : FAILURE in 1h 08m 18s","accounts_in_message":[],"_revision_number":1},{"id":"812f64d506e73c8fc891486f2ab71b182864cc0c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-01 23:01:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/1/check/gate-nova-docs-ubuntu-xenial/49bacbb//doc/build/html/ : SUCCESS in 2m 36s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-pep8-ubuntu-xenial/f3d4731/ : SUCCESS in 6m 05s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-python27-ubuntu-xenial/eb59a80/ : SUCCESS in 10m 43s\n- gate-nova-python35 http://logs.openstack.org/48/461448/1/check/gate-nova-python35/fbb07cb/ : SUCCESS in 10m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f7e9c8b/ : SUCCESS in 1h 11m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/dbc756d/ : SUCCESS in 47m 08s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/a1077d1/ : SUCCESS in 1h 00m 41s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-ubuntu-xenial/dbdffad/ : SUCCESS in 21m 35s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/8d0103f/ : SUCCESS in 26m 03s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/66848ab/ : SUCCESS in 40m 14s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/d66a7c6/ : SUCCESS in 37m 59s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/0197b90/ : SUCCESS in 42m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/86e817f/ : FAILURE in 1h 03m 32s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/84ece27/ : SUCCESS in 1h 11m 33s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/f01df1f/ : SUCCESS in 1h 09m 55s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a3f4426/ : FAILURE in 26m 48s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/785105d/ : SUCCESS in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa2db701d59f2b79e0a098ed5abff02d34a3850b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-02 01:01:56.000000000","message":"Patch Set 1:\n\nrecheck bug 1687511","accounts_in_message":[],"_revision_number":1},{"id":"6d1193aa2eca54f745e6e54665dc3197fdf9925d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-02 01:54:59.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/tempest-dsvm-full-xenial/975ca28/ : FAILURE in 52m 19s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/1/check/grenade-dsvm-xenial/3dbec6d/ : SUCCESS in 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0c4a8e32083661a337294abddb90fdd661ea0b5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-05-02 01:57:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/48/461448/1/check/ext-nova-zuul/dcecb12 : FAILURE in 30m 06s","accounts_in_message":[],"_revision_number":1},{"id":"aa30c8f924abdabede421245c737ceea27ceb45b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-02 02:11:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/1/ : FAILURE in 1h 08m 15s","accounts_in_message":[],"_revision_number":1},{"id":"630044bccad600676b1498125646ed09f5cc7a8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 02:21:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/1/check/gate-nova-docs-ubuntu-xenial/47ef4aa//doc/build/html/ : SUCCESS in 4m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-pep8-ubuntu-xenial/65d88c1/ : SUCCESS in 8m 35s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-python27-ubuntu-xenial/c09f4da/ : SUCCESS in 10m 14s\n- gate-nova-python35 http://logs.openstack.org/48/461448/1/check/gate-nova-python35/8e39ede/ : SUCCESS in 10m 56s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d4feec1/ : SUCCESS in 1h 04m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/43289b6/ : SUCCESS in 33m 38s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/71fb577/ : SUCCESS in 56m 27s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-ubuntu-xenial/a2e0be0/ : SUCCESS in 20m 10s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/9db9e61/ : SUCCESS in 24m 33s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/1961ae6/ : SUCCESS in 57m 44s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/52a2fbb/ : FAILURE in 35m 36s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/f50db10/ : SUCCESS in 44m 32s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/b069f4d/ : FAILURE in 1h 15m 39s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/aca475a/ : SUCCESS in 1h 14m 32s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/90961f9/ : SUCCESS in 1h 13m 37s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/1e581f8/ : FAILURE in 40m 11s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/046cf1a/ : SUCCESS in 38m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e83724d998aef3bd3681d6d590efebaa48db1617","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-05-02 03:22:41.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial: https://bugs.launchpad.net/bugs/1687511\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"2b5d52e950d5c9b877d4e2497917fef292be76e2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-05-02 04:22:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/48/461448/1/check-nova/Nova-ML2-Sriov/a4979af : SUCCESS in 53m 21s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/48/461448/1/check-nova/Nova-MACVTAP-ML2-Sriov/443d1af : SUCCESS in 52m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c88d6cff9016fe370fac7453600816d26e870dc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 14:45:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461448/1/check/tempest-dsvm-full-nfv-xenial/19eabb8 : SUCCESS in 2h 02m 28s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461448/1/check/tempest-dsvm-intel-nfv-xenial/9e25d6c : SUCCESS in 48m 03s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461448/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e95a231 : SUCCESS in 1h 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e09c15101a2ef52e97a07f4757d2057c3baa981c","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-02 16:29:26.000000000","message":"Patch Set 1:\n\nrecheck bug 1687511","accounts_in_message":[],"_revision_number":1},{"id":"549494bb66d87f62097fe6dbaf7a6c432faf72a8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-02 16:48:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"38bdf59c3d085c3e2a56cc6f34c657234c9c9636","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-05-02 17:35:22.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/461448/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"21401f5cdfcd536e3071eee1a877ef9c8d21e8ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-02 17:44:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/2/check/tempest-dsvm-full-xenial/1109fe3/ : SUCCESS in 51m 07s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/2/check/grenade-dsvm-xenial/598254d/ : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd494ea8b12fd0e506c244b06a844cd660ac2f6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 18:22:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/2/check/gate-nova-docs-ubuntu-xenial/5812510//doc/build/html/ : SUCCESS in 5m 46s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-pep8-ubuntu-xenial/5527e17/ : SUCCESS in 8m 24s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-python27-ubuntu-xenial/4c7a7de/ : SUCCESS in 12m 09s\n- gate-nova-python35 http://logs.openstack.org/48/461448/2/check/gate-nova-python35/8306f12/ : SUCCESS in 16m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/59c25f2/ : SUCCESS in 1h 10m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c33a024/ : SUCCESS in 36m 49s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/c7cc0cb/ : SUCCESS in 1h 08m 46s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-tox-functional-ubuntu-xenial/4230a71/ : SUCCESS in 20m 31s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/dea2091/ : SUCCESS in 24m 22s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/c915608/ : SUCCESS in 40m 42s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/c074566/ : FAILURE in 50m 34s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/aac7fd1/ : SUCCESS in 44m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/5707a8a/ : FAILURE in 1h 10m 48s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/d45510b/ : SUCCESS in 1h 26m 51s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/493cfb5/ : FAILURE in 1h 05m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/b4fbdb6/ : SUCCESS in 39m 47s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f580011/ : SUCCESS in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3230795f36802b250a4e70649d9dc0540b5358e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 19:05:48.000000000","message":"Patch Set 2:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461448/2/check-ibm-tempest-dsvm-full/dc984af : FAILURE in 1h 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"86116d7d0c62b5a967f6b1f03a6850437de111a7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-05-02 19:15:34.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/48/461448/2/check/dsvm-tempest-xen/e36b763 : FAILURE in 2h 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f994789123c9f1ad3ac8e201f476df38d1a8690e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-05-02 19:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/48/461448/2/check/dsvm-tempest-neutron-network/517b499 : SUCCESS in 2h 27m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fc6ce258cea7dd6da226a2fd2f04e57b5e26ab8c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-05-02 20:07:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/48/461448/2/check/ext-nova-zuul/fa78be4 : FAILURE in 42m 29s","accounts_in_message":[],"_revision_number":2},{"id":"bbe16dd4c9dc80f476689fbebe3fa81394112841","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-05-02 20:11:52.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-48-461448-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"306df6fbb0685fd96ad20bf856851fef45446548","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-02 20:54:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/2/ : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e94e7c2abb2f9966cae7d9ef4c4d3bf26dd6ec5e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-05-03 00:48:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/48/461448/2/check-nova/Nova-ML2-Sriov/55afb62 : SUCCESS in 52m 32s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/48/461448/2/check-nova/Nova-MACVTAP-ML2-Sriov/821c98e : SUCCESS in 45m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"a9ac08a2682596d97acd3604ab69462bf47a8ac9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-05-03 02:38:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-02/461448/2/check/tempest-dsvm-full-nfv-xenial/c6bef14 : SUCCESS in 2h 02m 44s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-02/461448/2/check/tempest-dsvm-intel-nfv-xenial/cb66be0 : SUCCESS in 47m 58s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-02/461448/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/3bbd6e8 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b16c1070e444d8afa2f5ce6fc368918aa99147a","author":{"_account_id":23314,"name":"lidong","email":"lidongbj@inspur.com","username":"lidong"},"date":"2017-05-03 02:52:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"49f139ddf1b848fc8aa654aa17566e3cbc52d3f4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-05-04 14:33:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16903a1a3975e2da99cfa2bfbfa25e20bb074b1d","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-05-04 15:36:48.000000000","message":"Patch Set 2:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":2},{"id":"1c1a63bdaac3dffcffcf482ce5837208f46a4c0d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-04 16:26:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/2/check/tempest-dsvm-full-xenial/307e1c2/ : SUCCESS in 49m 18s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/2/check/grenade-dsvm-xenial/53582e5/ : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdee6a808a3d2e329894a2ceec62f3ffa5fee50b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-04 17:07:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/2/check/gate-nova-docs-ubuntu-xenial/a5d5702//doc/build/html/ : SUCCESS in 4m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-pep8-ubuntu-xenial/081f8c5/ : SUCCESS in 6m 39s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-python27-ubuntu-xenial/bf65ff4/ : SUCCESS in 11m 38s\n- gate-nova-python35 http://logs.openstack.org/48/461448/2/check/gate-nova-python35/a763a8a/ : SUCCESS in 11m 09s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/3468990/ : SUCCESS in 1h 14m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/620bef9/ : SUCCESS in 51m 22s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/245e0d2/ : SUCCESS in 59m 24s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-tox-functional-ubuntu-xenial/b7f82ad/ : SUCCESS in 28m 16s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/860ba62/ : SUCCESS in 21m 51s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/925f489/ : SUCCESS in 40m 56s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/b8ca3cc/ : FAILURE in 40m 07s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/310fa17/ : SUCCESS in 44m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a63f195/ : SUCCESS in 1h 19m 28s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/484badd/ : SUCCESS in 1h 16m 40s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/98870d6/ : SUCCESS in 1h 14m 59s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/4fa716c/ : SUCCESS in 26m 21s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/fb4a795/ : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59b68724d7db1d87de83fe13cbac83b111dd319b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-04 17:17:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0aa272a94ff5b66e8dfaf16dea1e2a24e702a8a6","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-04 17:20:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/2/ : SUCCESS in 1h 42m 56s","accounts_in_message":[],"_revision_number":2},{"id":"187f6357eecd495a633a004ea31175791775234f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-04 17:23:38.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c723b0d9085f10e3ca85a84831a3868c11d72f8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-05-04 17:31:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/48/461448/2/check-vote/ext-nova-zuul/6dfc0cf : FAILURE in 1h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b610e45512b3e828e927c1960b8cdb938ccf0a60","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-04 17:55:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"99465f1079cfcd2c501639484c9361ebb1ed7d9d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-04 17:57:26.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"61a6d49f65f143dc3bcc494f97d7a07b582cd77b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-04 18:52:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/tempest-dsvm-full-xenial/444cec9/ : SUCCESS in 48m 22s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/grenade-dsvm-xenial/6e5d680/ : SUCCESS in 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"280093c0d7ef8d95b67e382497ca9a17b7442da1","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-05-04 19:08:36.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/48/461448/4/check/nova-out-of-tree-pvm/e601172 : SUCCESS in 1h 03m 37s\n- nova-in-tree-pvm http://184.172.12.213/48/461448/4/check/nova-in-tree-pvm/ce21437 : FAILURE in 54m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c1e3a3c8307699e1b5f3d5a86c7ab644e2259c4b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-04 19:26:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/4/check/gate-nova-docs-ubuntu-xenial/08605e2//doc/build/html/ : SUCCESS in 5m 11s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-pep8-ubuntu-xenial/90c83b7/ : SUCCESS in 8m 30s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-python27-ubuntu-xenial/30f17a3/ : SUCCESS in 12m 21s\n- gate-nova-python35 http://logs.openstack.org/48/461448/4/check/gate-nova-python35/c5cd8ca/ : SUCCESS in 12m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a96a5c2/ : FAILURE in 1h 00m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/fcc94b1/ : SUCCESS in 45m 16s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/dd946ae/ : SUCCESS in 1h 01m 52s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-tox-functional-ubuntu-xenial/86d5382/ : SUCCESS in 23m 54s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-tox-functional-py35-ubuntu-xenial/520d4c0/ : SUCCESS in 22m 30s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b248ced/ : SUCCESS in 45m 18s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/4b527bf/ : FAILURE in 57m 21s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/0cb162e/ : SUCCESS in 49m 29s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d8d0dde/ : SUCCESS in 1h 08m 09s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/48cae54/ : SUCCESS in 1h 16m 20s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/a918dd2/ : SUCCESS in 1h 08m 31s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8b8af11/ : SUCCESS in 25m 42s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/68cdb1f/ : SUCCESS in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d6a1e324fdb8d393e0bb47896bddef1b355861b","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-04 19:56:30.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/4/ : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ba21db1bfa6fae328ab5ad9f08e72e9f844e815d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-05-04 20:18:33.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/461448/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"2e370470dc8e0419bb58d052f2d66cd695e278f5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-05-04 20:21:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/48/461448/4/check/dsvm-tempest-neutron-network/45e3675 : SUCCESS in 2h 02m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f56967caa057d3d45673eae5fb69ec78baba28fb","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-05-04 20:54:58.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/48/461448/4/check/dsvm-tempest-xen/11db599 : FAILURE in 2h 14m 33s","accounts_in_message":[],"_revision_number":4},{"id":"91d1c5fa0f3fd9c84406167ee16da39a62ebe109","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-05-04 21:18:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/48/461448/4/check/check-dsvm-tempest-vz7-exe-minimal/7126c0a : SUCCESS in 1h 17m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"8178df771fa08da07b507b4abc9b913cdd56702d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-05-04 22:23:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-04/461448/4/check/tempest-dsvm-full-nfv-xenial/07c5a08 : SUCCESS in 2h 06m 16s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-04/461448/4/check/tempest-dsvm-intel-nfv-xenial/a1dac13 : SUCCESS in 51m 09s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-04/461448/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/5009918 : FAILURE in 1h 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b645e9e9c514c0fa34c740947bf353cbc3469adb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-05-04 23:15:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/48/461448/4/check-nova/Nova-ML2-Sriov/5b8bc80 : SUCCESS in 44m 49s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/48/461448/4/check-nova/Nova-MACVTAP-ML2-Sriov/9cae95b : SUCCESS in 50m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"d61a31dc026753a5f73033ed51ca41a53ed1252c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-05 00:24:51.000000000","message":"Patch Set 4:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461448/4/check-ibm-tempest-dsvm-full/54330f1 : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b8ba4de51776956d6974409025edd2a43b3c12c6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-05-05 13:19:24.000000000","message":"Patch Set 4:\n\nrecheck bug 1687044","accounts_in_message":[],"_revision_number":4},{"id":"403f4f2482da83d81afaebb90ca4019c87d8c880","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-05 14:15:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/tempest-dsvm-full-xenial/9e7ec8c/ : SUCCESS in 54m 55s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/grenade-dsvm-xenial/d6d38f8/ : SUCCESS in 40m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c820f43589520f9ecda5fc26042b8399a36ccbe","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-05-05 14:45:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9759f21f321de6a21889f78a7adaf7c89044419a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2017-05-05 14:46:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a48b7612d77efd7a42db0025af04207620f0d7c0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-05 15:08:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/4/check/gate-nova-docs-ubuntu-xenial/937fec0//doc/build/html/ : SUCCESS in 4m 46s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-pep8-ubuntu-xenial/7d178e5/ : SUCCESS in 8m 33s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-python27-ubuntu-xenial/ffe5c82/ : SUCCESS in 15m 27s\n- gate-nova-python35 http://logs.openstack.org/48/461448/4/check/gate-nova-python35/56080b5/ : SUCCESS in 13m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/74ce416/ : SUCCESS in 1h 20m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/6c93200/ : SUCCESS in 58m 53s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/8266d90/ : SUCCESS in 1h 24m 53s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-tox-functional-ubuntu-xenial/8c5a263/ : SUCCESS in 28m 11s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-nova-tox-functional-py35-ubuntu-xenial/d52a552/ : SUCCESS in 27m 08s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/e80fd15/ : SUCCESS in 43m 09s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/48/461448/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a618ecd/ : FAILURE in 40m 01s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/5ab001d/ : SUCCESS in 42m 42s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/49bd5f7/ : SUCCESS in 1h 15m 18s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/8fd2424/ : SUCCESS in 1h 20m 11s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/095bded/ : SUCCESS in 1h 46m 15s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d78fb24/ : SUCCESS in 26m 23s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/48/461448/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/30a66e2/ : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f2de80cbbbf17c243f1ea8725064b74c5d3659c","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-05 15:12:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461448/4/ : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":4},{"id":"13b1c931305cf167865a6f0193969847702da9ee","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-05-05 15:24:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.69.151.247/48/461448/4/check-nova/Nova-ML2-Sriov/c8fd76f : SUCCESS in 46m 00s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.69.151.247/48/461448/4/check-nova/Nova-MACVTAP-ML2-Sriov/427b345 : SUCCESS in 39m 27s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"dedd180a2c6a7825ee9a13dddb0c924949f624e6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-05-05 15:27:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-05/461448/4/check/tempest-dsvm-full-nfv-xenial/45491d9 : SUCCESS in 2h 07m 34s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-05/461448/4/check/tempest-dsvm-intel-nfv-xenial/35afa3a : SUCCESS in 52m 19s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-05/461448/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/97cb4ae : SUCCESS in 1h 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6984c275afacd9fad2d5c06f470952ba0b3d84b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-05-05 15:35:10.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/tempest-dsvm-full-xenial/29d0e29/ : SUCCESS in 48m 24s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/48/461448/4/check/grenade-dsvm-xenial/e4dade0/ : FAILURE in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e9375990115cf0aef28da7da1f566d4527a0fa32","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-05-05 15:52:35.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/48/461448/4/check/nova-out-of-tree-pvm/90313ca : SUCCESS in 1h 05m 53s\n- nova-in-tree-pvm http://184.172.12.213/48/461448/4/check/nova-in-tree-pvm/24219ba : FAILURE in 35m 08s","accounts_in_message":[],"_revision_number":4},{"id":"e8408af9c48f9d09e66cfe3894cdd95a685a4a2a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-05-05 17:01:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/48/461448/4/check/check-dsvm-tempest-vz7-exe-minimal/353b805 : SUCCESS in 1h 19m 03s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"97dfb37550fe82a6ff3a0e381234ecc232b2d122","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-05-05 19:41:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/48/461448/4/check-vote/ext-nova-zuul/6ec738d : FAILURE in 1h 00m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c5ed64432f783b91da53a8960962a27befc0770e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-05-07 03:23:32.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-48-461448-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"762caffe64c7bc480dc0cfdc29f44b5b9fd3d0e5","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2017-05-10 22:33:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d38e3a3bcf6797f3b2ff3568c9518b9582625208","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-05-11 05:07:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3b5560bdff07b9f872ac6d5850c9204f1c6b33fc","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2017-05-15 09:22:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8752034ede253b18c055d91622935aeace31c430","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-15 09:22:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"760f714b8019e60a7fb9f92202fdf9f11aeaab30","date":"2017-05-15 10:38:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"77a0f3c219fc317b68442912847ca0f1683b44d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-15 10:38:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/48/461448/4/gate/gate-nova-docs-ubuntu-xenial/05e7209//doc/build/html/ : SUCCESS in 4m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-nova-pep8-ubuntu-xenial/d792d08/ : SUCCESS in 9m 55s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-nova-python27-ubuntu-xenial/50c2e4c/ : SUCCESS in 13m 41s\n- gate-nova-python35 http://logs.openstack.org/48/461448/4/gate/gate-nova-python35/64462e0/ : SUCCESS in 11m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4474bec/ : SUCCESS in 1h 14m 42s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/c9370bf/ : SUCCESS in 49m 35s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-tempest-dsvm-py35-ubuntu-xenial/1de68d1/ : SUCCESS in 1h 09m 10s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-nova-tox-functional-ubuntu-xenial/b4bd0b4/ : SUCCESS in 20m 39s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-nova-tox-functional-py35-ubuntu-xenial/c1f805a/ : SUCCESS in 21m 56s\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-tempest-dsvm-cells-ubuntu-xenial/065312e/ : SUCCESS in 55m 20s\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/48/461448/4/gate/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5135864/ : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbfc7ac3c4de82e8555699fd045e4253972d14de","revisions":{"91967719700b6ed3c087720691037332f052be5e":{"kind":"REWORK","_number":1,"created":"2017-05-01 16:37:14.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/48/461448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/48/461448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/48/461448/1"}}},"commit":{"parents":[{"commit":"e7565dbd51715dece30c405e12dc797a4409cc7c","subject":"db api: add compute_node_get_by_uuid method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7565dbd51715dece30c405e12dc797a4409cc7c"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-01 16:36:02.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-01 16:36:02.000000000","tz":-240},"subject":"objects: add ComputeNode.get_by_uuid method","message":"objects: add ComputeNode.get_by_uuid method\n\nThe HostAPI is going to use this to uniquely identify a host\nacross cells.\n\nPart of blueprint service-hyper-uuid-in-api\n\nChange-Id: I21f4b18e4554e9e912496e623461129196c72852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91967719700b6ed3c087720691037332f052be5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91967719700b6ed3c087720691037332f052be5e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e7565dbd51715dece30c405e12dc797a4409cc7c","is_merged_in_target_branch":false,"change_id":"Ica92e9ca7da81798dee4f7829670c1f85df442da","change_number":461447,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"f7e1c1cdab2fff80e54cb30c104500168bfa9b29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-05-02 16:48:54.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/48/461448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/48/461448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/48/461448/2"}}},"commit":{"parents":[{"commit":"8433467c3770fa2dbdeff25dde4d42b53e8a3e30","subject":"db api: add compute_node_get_by_uuid method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8433467c3770fa2dbdeff25dde4d42b53e8a3e30"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-01 16:36:02.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-02 16:34:13.000000000","tz":-240},"subject":"objects: add ComputeNode.get_by_uuid method","message":"objects: add ComputeNode.get_by_uuid method\n\nThe HostAPI is going to use this to uniquely identify a host\nacross cells.\n\nPart of blueprint service-hyper-uuid-in-api\n\nChange-Id: I21f4b18e4554e9e912496e623461129196c72852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7e1c1cdab2fff80e54cb30c104500168bfa9b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7e1c1cdab2fff80e54cb30c104500168bfa9b29"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8433467c3770fa2dbdeff25dde4d42b53e8a3e30","is_merged_in_target_branch":false,"change_id":"Ica92e9ca7da81798dee4f7829670c1f85df442da","change_number":461447,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"21af728552e5bcee6411ab2a23212ad46a70837d":{"kind":"REWORK","_number":3,"created":"2017-05-04 17:55:08.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/48/461448/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/48/461448/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/48/461448/3"}}},"commit":{"parents":[{"commit":"6e73eb4d0d38f84d44e5d9d97d876d52702e3a78","subject":"objects: add Service.get_by_uuid method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e73eb4d0d38f84d44e5d9d97d876d52702e3a78"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-01 16:36:02.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-04 17:54:35.000000000","tz":-240},"subject":"objects: add ComputeNode.get_by_uuid method","message":"objects: add ComputeNode.get_by_uuid method\n\nThe HostAPI is going to use this to uniquely identify a host\nacross cells.\n\nPart of blueprint service-hyper-uuid-in-api\n\nChange-Id: I21f4b18e4554e9e912496e623461129196c72852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21af728552e5bcee6411ab2a23212ad46a70837d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21af728552e5bcee6411ab2a23212ad46a70837d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6e73eb4d0d38f84d44e5d9d97d876d52702e3a78","is_merged_in_target_branch":true,"change_id":"I66ed343dd02099b2d4a9e3ff7c496b09c0822ce6","change_number":461429,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fbfc7ac3c4de82e8555699fd045e4253972d14de":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2017-05-04 17:57:26.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/48/461448/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/48/461448/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/48/461448/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/48/461448/4"}}},"commit":{"parents":[{"commit":"6e73eb4d0d38f84d44e5d9d97d876d52702e3a78","subject":"objects: add Service.get_by_uuid method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e73eb4d0d38f84d44e5d9d97d876d52702e3a78"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-01 16:36:02.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-05-04 17:57:02.000000000","tz":-240},"subject":"objects: add ComputeNode.get_by_uuid method","message":"objects: add ComputeNode.get_by_uuid method\n\nThe HostAPI is going to use this to uniquely identify a host\nacross cells.\n\nDepends-On: I940979d85fc105c9733ae19c406e85debe8fdffb\n\nPart of blueprint service-hyper-uuid-in-api\n\nChange-Id: I21f4b18e4554e9e912496e623461129196c72852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbfc7ac3c4de82e8555699fd045e4253972d14de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbfc7ac3c4de82e8555699fd045e4253972d14de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6e73eb4d0d38f84d44e5d9d97d876d52702e3a78","is_merged_in_target_branch":true,"change_id":"I66ed343dd02099b2d4a9e3ff7c496b09c0822ce6","change_number":461429,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
