)]}'
{"id":"openstack%2Fnova~461631","triplet_id":"openstack%2Fnova~master~Ic3a634071b5b7629151f434c8549031b12f9baea","project":"openstack/nova","branch":"master","topic":"fix_file_mode","hashtags":[],"change_id":"Ic3a634071b5b7629151f434c8549031b12f9baea","subject":"Remove unnecessary execute permissions","status":"MERGED","created":"2017-05-02 05:40:30.000000000","updated":"2017-05-03 11:47:51.000000000","submitted":"2017-05-03 11:47:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"71e00a990f6f91c0af251ecec69911990c596e86","_number":461631,"virtual_id_number":461631,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-05-03 11:47:51.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-05-03 11:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2017-05-02 15:04:08.000000000","_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-05-03 11:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-02 15:04:08.000000000","updated_by":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"reviewer":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"state":"REVIEWER"},{"updated":"2017-05-03 11:11:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2017-05-03 11:47:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"d5d10bc30a239674ed11f2b5084d4477f720bec8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-05-02 05:40:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55d694787c60a9420bcb9687854c9f1cbe20a946","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 05:50:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/31/461631/1/check/gate-nova-docs-ubuntu-xenial/f4db8b1//doc/build/html/ : SUCCESS in 2m 27s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/31/461631/1/check/gate-nova-pep8-ubuntu-xenial/98b79e2/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"314ba855975c756a862d7b6e69dd2206f99e1e07","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-05-02 06:16:34.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-31-461631-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"f8e29236dcf692cb33bd358e490c345877f3d3f5","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-05-02 06:41:20.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/31/461631/1/check/nova-out-of-tree-pvm/565138b : FAILURE in 25m 13s\n- nova-in-tree-pvm http://184.172.12.213/31/461631/1/check/nova-in-tree-pvm/40d6c31 : FAILURE in 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0c3353eccb47dfcde8fa7216950765151fb729d8","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-02 06:48:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461631/1/ : FAILURE in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9c82901f7d5642398cb7a932ea997d71be2b37b3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-05-02 06:51:01.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/461631/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8e9623ce9e1d45439821b508d8b8f837fc873d62","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 08:03:26.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461631/1/check-ibm-tempest-dsvm-full/d73465f : FAILURE in 2h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"02502a8a2f74b9dc8bdfb6a5440131fa506c9910","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-05-02 08:31:39.000000000","message":"Patch Set 1:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":1},{"id":"7bbb5fbcbef61a2df17eb6889ea0fece1caa1579","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 08:39:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/31/461631/1/check/gate-nova-docs-ubuntu-xenial/4cd767e//doc/build/html/ : SUCCESS in 2m 25s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/31/461631/1/check/gate-nova-pep8-ubuntu-xenial/3fed7bd/ : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4d82c4a86f7d18787afe7e13f4abdf453cbc1689","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 10:13:50.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461631/1/check-ibm-tempest-dsvm-full/69b93bb : FAILURE in 1h 41m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c062c4a8137839b8d883671593f3732375155cba","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-02 14:59:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461631/1/ : SUCCESS in 1h 46m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fd853932f75f33690fe5285353cbedb433806fc5","author":{"_account_id":679,"name":"Kevin L. Mitchell","email":"klmitch@mit.edu","username":"klmitch"},"date":"2017-05-02 15:04:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fb664af08be489a3c503b423cd66725bb347f74","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-05-02 15:21:58.000000000","message":"Patch Set 1:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":1},{"id":"f7853bcfd4c7a5c787377321442e73aba989f662","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-02 15:36:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/31/461631/1/check/gate-nova-docs-ubuntu-xenial/b6e1742//doc/build/html/ : SUCCESS in 3m 47s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/31/461631/1/check/gate-nova-pep8-ubuntu-xenial/f6924e5/ : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f11153bc460c5d953e23fbde1bf64d36ed7b11d7","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 17:43:04.000000000","message":"Patch Set 1:\n\nTesting completed on KVM for IBM z Systems. For rechecking only on the KVM for IBM z Systems CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-ibm-tempest-dsvm-full http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/461631/1/check-ibm-tempest-dsvm-full/1940353 : SUCCESS in 1h 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"77b7553716d1b9903b50d751672d5bd0907e840f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-05-02 19:16:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461631/1/check/tempest-dsvm-full-nfv-xenial/c0bb2c6 : SUCCESS in 1h 55m 46s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461631/1/check/tempest-dsvm-intel-nfv-xenial/4443ea8 : SUCCESS in 47m 59s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-05-01/461631/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/86bc407 : SUCCESS in 1h 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df2410c472200e90b176147406791e5b54b7467c","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-05-02 19:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/461631/1/ : SUCCESS in 1h 42m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a3efd12f673e2245a3983ba32f1f0ed6e602ba44","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-05-03 11:11:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivialFix indeedd","accounts_in_message":[],"_revision_number":1},{"id":"42baaedc29e659886d9c48a69b3f98b0a7dd8127","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-03 11:11:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"084850ffa43c8cdf5c9cb4091fe93a9d2da1d2c7","date":"2017-05-03 11:47:51.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"61cf76654cc0d8fda61edb7d52cd09d4d9bfbbfb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-03 11:47:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/31/461631/1/gate/gate-nova-docs-ubuntu-xenial/81e1fc3//doc/build/html/ : SUCCESS in 5m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/31/461631/1/gate/gate-nova-pep8-ubuntu-xenial/4a806c9/ : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dccd839dfd97501fb5ad3e85b6418e4ecf7064b4","revisions":{"dccd839dfd97501fb5ad3e85b6418e4ecf7064b4":{"kind":"REWORK","_number":1,"created":"2017-05-02 05:40:30.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/31/461631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/461631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/461631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/461631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/461631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/461631/1"}}},"commit":{"parents":[{"commit":"d6d5c6be0cc0738bb7d67ca8391e5cc4f0a419f2","subject":"Merge \"Remove nova-cert\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6d5c6be0cc0738bb7d67ca8391e5cc4f0a419f2"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-05-02 05:35:47.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-05-02 05:39:25.000000000","tz":540},"subject":"Remove unnecessary execute permissions","message":"Remove unnecessary execute permissions\n\nTrivialFix\nChange-Id: Ic3a634071b5b7629151f434c8549031b12f9baea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dccd839dfd97501fb5ad3e85b6418e4ecf7064b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dccd839dfd97501fb5ad3e85b6418e4ecf7064b4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d6d5c6be0cc0738bb7d67ca8391e5cc4f0a419f2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
