)]}'
{"id":"openstack%2Fnova~480957","triplet_id":"openstack%2Fnova~master~I044014861a6d7ad3f81c94b44fff6b30dbd6d589","project":"openstack/nova","branch":"master","topic":"bp/nested-resource-providers","hashtags":[],"change_id":"I044014861a6d7ad3f81c94b44fff6b30dbd6d589","subject":"placement: add ProviderTree.is_inventory_empty","status":"ABANDONED","created":"2017-07-06 10:20:28.000000000","updated":"2017-12-13 14:44:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"584c3e1340380864dff69feccb8bf18a9e4e8f88","_number":480957,"virtual_id_number":480957,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":-1,"date":"2017-10-09 14:52:03.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2017-10-09 14:49:45.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2017-10-10 00:45:36.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":-1,"date":"2017-10-09 14:49:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2017-10-23 05:50:46.000000000","_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2017-10-09 14:50:41.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2017-10-09 14:51:42.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":-1,"date":"2017-10-09 14:48:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-06 11:29:12.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-07-06 11:35:15.000000000","updated_by":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"reviewer":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"state":"REVIEWER"},{"updated":"2017-07-06 13:02:00.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-22 15:24:58.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-22 16:01:43.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-22 16:08:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-22 17:45:17.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-09-22 19:26:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-23 07:42:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 17:15:47.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2017-10-09 14:49:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 14:49:45.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 14:50:41.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 14:51:42.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 14:52:03.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-10 00:45:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 05:50:46.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"}],"messages":[{"id":"82a0d1523fbaf6e72ed6c9c6d1002a715b2ce94f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-07-06 10:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbe593a5ca20372f5cc8054fb40d652c0b0fd323","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-07-06 10:34:41.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x : FAILURE in 2m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c79aa81e5bd876ddf2a22b1bb9df70f1a05c5886","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-07-06 11:24:01.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/480957/1/check/tempest-dsvm-full-xenial/7cc028b/ : FAILURE in 56m 24s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/480957/1/check/grenade-dsvm-xenial/81c82b7/ : SUCCESS in 44m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee057268bf3c0ee1bd24032deda75cbb90fdb0c7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-07-06 11:29:12.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/480957/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"59a60e03052d55602c039e69306446afa918d0a8","author":{"_account_id":20040,"name":"Microsoft Hyper-V Compute CI","email":"microsoft_nova_ci@microsoft.com","username":"microsoft_nova_ci"},"date":"2017-07-06 11:35:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Microsoft Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dvsm-full-tempest http://64.119.130.115/nova/480957/1/ : FAILURE in 1h 08m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d9684da0e49b32439e4c1d40689f30355cd1d9b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-06 12:17:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/57/480957/1/check/gate-nova-docs-ubuntu-xenial/f5fcc3c//doc/build/html/ : SUCCESS in 6m 31s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-nova-pep8-ubuntu-xenial/b592d60/ : SUCCESS in 7m 27s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-nova-python27-ubuntu-xenial/901e7ef/ : SUCCESS in 11m 36s\n- gate-nova-python35 http://logs.openstack.org/57/480957/1/check/gate-nova-python35/12e1c5a/ : SUCCESS in 11m 09s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/96b4f8f/ : SUCCESS in 1h 31m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c1a1c40/ : SUCCESS in 34m 36s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/b97120a/ : SUCCESS in 1h 09m 19s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-nova-tox-functional-ubuntu-xenial/df9f8bf/ : FAILURE in 26m 36s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/11b7a37/ : FAILURE in 23m 34s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0e48ba6/ : SUCCESS in 56m 58s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/57/480957/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/ec10c15/ : SUCCESS in 51m 49s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/e4eb01a/ : SUCCESS in 48m 58s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/9a8f04e/ : SUCCESS in 1h 18m 58s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/b5ceaad/ : SUCCESS in 1h 36m 50s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/e585853/ : SUCCESS in 1h 19m 35s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/27022eb/ : SUCCESS in 39m 14s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/57/480957/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/12026b3/ : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37896f87fcbb34846be24c403c40bbecb6bcb475","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-07-06 13:01:37.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/480957/1/check/nova-out-of-tree-pvm/f38598f : FAILURE in 2h 38m 34s\n- nova-in-tree-pvm http://184.172.12.213/57/480957/1/check/nova-in-tree-pvm/4525b03 : FAILURE in 1h 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a757bdd4101de065f1e05361132c37974bd2ca69","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-07-06 13:02:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-06/480957/1/check/tempest-dsvm-full-nfv-xenial/2a11fbe : SUCCESS in 2h 16m 30s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-06/480957/1/check/tempest-dsvm-intel-nfv-xenial/6780366 : SUCCESS in 48m 21s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-07-06/480957/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/6703e3a : SUCCESS in 1h 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e11b65459b3344739ce3dc6189089abc96c4f9cd","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-07-06 13:57:46.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/480957/1/check/dsvm-tempest-xen/3d2d56a : FAILURE in 2h 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f194cccd35ef9b85eb78a1a5f1b6468f8c1f0141","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-07-06 14:51:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/480957/1/check/dsvm-tempest-neutron-network/373e70c : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"629b8a21d43d7f3b99d0f71db087d6196e7028a6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-07-06 15:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/57/480957/1/check-nova/Nova-ML2-Sriov/17171ea : SUCCESS in 57m 13s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/57/480957/1/check-nova/Nova-MACVTAP-ML2-Sriov/186dda8 : FAILURE in 29m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"f04b9d0ff5109f7fb6c8593825a3d5403b15b102","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-07-06 15:30:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://52.169.200.208/57/480957/1/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/5cfbaf6 : SUCCESS in 1h 05m 00s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"255986a7630e0c1d65a49c8604de7d3d8b886642","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-22 14:26:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"14ec950b2d5ce34f8ab8e7471c13d84fcfb88f99","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-22 15:24:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/480957/2/check/tempest-dsvm-full-xenial/fb028f1/ : SUCCESS in 57m 30s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/480957/2/check/grenade-dsvm-xenial/8835a4b/ : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d072f6da59691fb45b4d6899671c28ea4fc6d478","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-22 15:25:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"1ac16b75bd5689f7cf8934cf115ce6c90b170595","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-22 16:01:43.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/480957/2/check/nova-out-of-tree-pvm/c1af5cd : FAILURE in 32m 22s\n- nova-in-tree-pvm http://184.172.12.213/57/480957/2/check/nova-in-tree-pvm/b5249ba : FAILURE in 1h 34m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9aece9a5b8c34d49d7b6aea71741beced369b3dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 16:07:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/57/480957/2/check/gate-nova-docs-ubuntu-xenial/ec33103//doc/build/html/ : SUCCESS in 4m 48s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-nova-pep8-ubuntu-xenial/a1261d0/ : SUCCESS in 9m 29s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-nova-python27-ubuntu-xenial/5ec05d8/ : SUCCESS in 11m 16s\n- gate-nova-python35 http://logs.openstack.org/57/480957/2/check/gate-nova-python35/30ff976/ : SUCCESS in 11m 39s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c871f45/ : SUCCESS in 1h 38m 29s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/651ba31/ : SUCCESS in 53m 52s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/555d58a/ : SUCCESS in 1h 26m 40s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-nova-tox-functional-ubuntu-xenial/b0d5a7b/ : SUCCESS in 20m 58s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/9f5f6a4/ : SUCCESS in 17m 39s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8b1e176/ : SUCCESS in 1h 02m 06s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/57/480957/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/5b43222/ : FAILURE in 56m 30s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/e0f71ef/ : SUCCESS in 1h 15m 22s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7f1a3ef/ : SUCCESS in 1h 02m 27s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/6401108/ : SUCCESS in 1h 22m 08s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/bf10b06/ : SUCCESS in 1h 17m 29s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5508c96/ : SUCCESS in 39m 24s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/57/480957/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c379aa3/ : SUCCESS in 43m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e59ed454a20bd901af48da4dbccb0f4ffdc763e6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-22 16:08:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"71fa2b9179749b733678068467a6b40fd601863d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-22 16:25:52.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/57/480957/2/check/dsvm-tempest-neutron-network/f00ca77 : FAILURE in 1h 57m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cc828694a4bad9474898bab84d70de03e225bbdb","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-22 16:26:30.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/480957/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e5360d1 : FAILURE in 1h 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5037a040694f28e689ab60a16ba4260251336479","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-22 16:47:27.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/57/480957/2/check/dsvm-tempest-xen/783cd15 : FAILURE in 2h 17m 14s","accounts_in_message":[],"_revision_number":2},{"id":"108e57a10693217c88d8b0b61560b1a55ec316ee","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-22 17:45:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/480957/2/ : SUCCESS in 2h 44m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5488ae6da764325db42bea1cc432455aaeaa9810","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-22 19:26:18.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-480957-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"a4a7e3564336787961574cac5d97684b8712201e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-23 07:42:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/57/480957/2/check-vote/ext-nova-zuul/7bdd4b7 : FAILURE in 1h 15m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e444cb56e7620ffdd8a2cdf2b8aa10549f4437ba","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-04 17:15:47.000000000","message":"Patch Set 2: Code-Review-1\n\n+1 on the concept.\n\nNeeds UT.","accounts_in_message":[],"_revision_number":2},{"id":"99d745fc56142d733cd4853448a96b2bc7ec56a3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 19:04:11.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ce024a4aad3ca5baac632dc3ba9edb235f971bdd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-09 14:48:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84b3b0e8e7a62df23eabac98f182f4908ff907a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-09 14:49:02.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c5ad91bc2e2856fb277b6dd1697460798cf27220","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 14:49:45.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"cabcb2bc58ebf0e443551fa3f41dd8b5ab680192","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-09 14:50:41.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"41b6d5730a0c1c6a147c19bf8f0ece2418ff4dad","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-09 14:51:42.000000000","message":"Patch Set 3:\n\nMerge failed. To try again use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":3},{"id":"5ab061e9a59fa192ca0aa2ab35e3651b12c06f99","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-09 14:52:03.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"f373a847b7d0cc345997d29a9bd161ed5b25af8b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-10 00:45:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ba7da20b9d2f0742200c8cc764e0ac83c81e82c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-12-13 14:44:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40808eb182452a5a78a7ae75e5fe83069347a4a9","revisions":{"ab6fe729272b5c852086acbc6dd41e0964f76261":{"kind":"REWORK","_number":1,"created":"2017-07-06 10:20:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/480957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/480957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/480957/1"}}},"commit":{"parents":[{"commit":"4b4f25d9b484ff1b3c545343d0a0622ec6b41a87","subject":"placement: update client to set parent provider","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b4f25d9b484ff1b3c545343d0a0622ec6b41a87"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-07-06 09:05:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-07-06 10:20:10.000000000","tz":60},"subject":"placement: add ProviderTree.is_inventory_empty","message":"placement: add ProviderTree.is_inventory_empty\n\nStandardize how we check for resource provider \"emptiness\".\n\nChange-Id: I044014861a6d7ad3f81c94b44fff6b30dbd6d589\nblueprint: nested-resource-providers\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab6fe729272b5c852086acbc6dd41e0964f76261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab6fe729272b5c852086acbc6dd41e0964f76261"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4b4f25d9b484ff1b3c545343d0a0622ec6b41a87","is_merged_in_target_branch":false,"change_id":"I2f367b06e683ed7c815dd9e0536a46e5f0a27e6c","change_number":385693,"patch_set_number":40,"change_status":"MERGED"}],"branch":"refs/heads/master"},"25220d2eecf3e09fe400caea312ab8b6f127d855":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-09-22 14:26:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/480957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/480957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/480957/2"}}},"commit":{"parents":[{"commit":"1cbcee55ae27e55b1cd27a1f7132966a55dd6c20","subject":"placement: update client to set parent provider","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cbcee55ae27e55b1cd27a1f7132966a55dd6c20"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-07-06 09:05:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-22 14:23:21.000000000","tz":60},"subject":"placement: add ProviderTree.is_inventory_empty","message":"placement: add ProviderTree.is_inventory_empty\n\nStandardize how we check for resource provider \"emptiness\".\n\nChange-Id: I044014861a6d7ad3f81c94b44fff6b30dbd6d589\nblueprint: nested-resource-providers\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25220d2eecf3e09fe400caea312ab8b6f127d855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25220d2eecf3e09fe400caea312ab8b6f127d855"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1cbcee55ae27e55b1cd27a1f7132966a55dd6c20","is_merged_in_target_branch":false,"change_id":"I2f367b06e683ed7c815dd9e0536a46e5f0a27e6c","change_number":385693,"patch_set_number":41,"change_status":"MERGED"}],"branch":"refs/heads/master"},"40808eb182452a5a78a7ae75e5fe83069347a4a9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-10-09 14:48:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/480957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/480957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/480957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/480957/3"}}},"commit":{"parents":[{"commit":"6f394e3ed1fcd0b54e9df152ab600f80131caae0","subject":"placement: update client to set parent provider","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f394e3ed1fcd0b54e9df152ab600f80131caae0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-07-06 09:05:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2017-10-09 14:48:44.000000000","tz":0},"subject":"placement: add ProviderTree.is_inventory_empty","message":"placement: add ProviderTree.is_inventory_empty\n\nStandardize how we check for resource provider \"emptiness\".\n\nChange-Id: I044014861a6d7ad3f81c94b44fff6b30dbd6d589\nblueprint: nested-resource-providers\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40808eb182452a5a78a7ae75e5fe83069347a4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40808eb182452a5a78a7ae75e5fe83069347a4a9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6f394e3ed1fcd0b54e9df152ab600f80131caae0","is_merged_in_target_branch":false,"change_id":"I2f367b06e683ed7c815dd9e0536a46e5f0a27e6c","change_number":385693,"patch_set_number":43,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
