)]}'
{"id":"openstack%2Fnova~493865","triplet_id":"openstack%2Fnova~master~I40a244601e1612a87bd41cd6ad652202035a1ce4","project":"openstack/nova","branch":"master","hashtags":[],"change_id":"I40a244601e1612a87bd41cd6ad652202035a1ce4","subject":"cover migration cases with functional tests","status":"MERGED","created":"2017-08-15 13:01:03.000000000","updated":"2018-07-06 00:55:09.000000000","submitted":"2018-07-06 00:55:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"493865-1530838509791-1c2e32f6","meta_rev_id":"7db89b8fbd62a6a907b4f45e8b600c9b06407f33","_number":493865,"virtual_id_number":493865,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-07-05 00:23:08.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"date":"2018-07-03 14:35:54.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"date":"2018-07-03 15:26:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-07-06 00:55:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2018-07-03 15:48:45.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-07-04 13:07:03.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2018-07-04 03:19:50.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2018-07-04 10:05:28.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"recommended":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2018-07-04 07:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2018-07-05 23:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2018-07-03 22:22:11.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2018-07-05 23:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-29 13:05:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2017-09-19 09:28:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2017-09-25 15:54:51.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2017-09-26 11:24:37.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-26 13:28:29.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 09:49:22.000000000","updated_by":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"reviewer":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"state":"REVIEWER"},{"updated":"2017-11-25 08:09:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-27 16:54:29.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2017-12-04 12:47:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-04-03 21:04:20.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-05-10 02:25:20.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-05-10 02:51:58.000000000","updated_by":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"reviewer":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"state":"REVIEWER"},{"updated":"2018-07-03 14:35:54.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-03 15:26:26.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-07-03 15:48:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-03 22:22:11.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2018-07-04 03:19:50.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-07-04 07:35:59.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2018-07-04 10:05:28.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-04 13:07:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-05 00:23:08.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-05 23:24:10.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-07-06 00:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fa988e87f6e19f4a9f28fba31a4b8716d8d8003","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-08-15 13:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"124b4ba901f9af45e6aa43cf5e3b356a84f6d0e0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-15 13:26:42.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/493865/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"15503deb5fcf0df2ce37c9f4aa5d19a5c632a142","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-15 13:27:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/1/check/gate-nova-docs-ubuntu-xenial/0561017//doc/build/html/ : SUCCESS in 4m 47s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/1/check/gate-nova-pep8-ubuntu-xenial/d8f7012/ : SUCCESS in 8m 50s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/1/check/gate-nova-python27-ubuntu-xenial/5d6e893/ : SUCCESS in 13m 01s\n- gate-nova-python35 http://logs.openstack.org/65/493865/1/check/gate-nova-python35/a6e62f2/ : SUCCESS in 12m 34s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/1/check/gate-nova-tox-functional-ubuntu-xenial/b096ca9/ : SUCCESS in 23m 40s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/c2dbce6/ : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a65af380013a3c177246907782175b928e9f1bda","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-15 14:37:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/1/check/nova-out-of-tree-pvm/7feff8e : SUCCESS in 1h 34m 25s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/1/check/nova-in-tree-pvm/f0b8cdb : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"137c2bd0352ae81de3ed8bbb36e5bc4ad81b3278","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-08-15 14:39:09.000000000","message":"Patch Set 1: Code-Review+1\n\nmoar test, yay","accounts_in_message":[],"_revision_number":1},{"id":"998ebdf58459312b578254fa67e0ef9e768a22ba","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-15 14:46:29.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"14734f09f511878c9bacf1480460e72ef74db27f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-15 14:58:12.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/85632d1 : SUCCESS in 1h 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0b897791a8a540fbf19e83ffbaed02bba469eb17","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-15 15:16:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-15/493865/1/check/tempest-dsvm-full-nfv-xenial/119a966 : SUCCESS in 2h 14m 49s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-15/493865/1/check/tempest-dsvm-intel-nfv-xenial/461bf7a : SUCCESS in 49m 23s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-15/493865/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/f6695e9 : SUCCESS in 1h 36m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fa9b00f942c9304b8db8a156c887ca6de094e83","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-15 15:22:57.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/493865/1/ : FAILURE in 2h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"93b1e44bfff160af15c2691fc9266e8582caf459","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-08-16 15:36:49.000000000","message":"Patch Set 1: Code-Review+1\n\nnice test.","accounts_in_message":[],"_revision_number":1},{"id":"f84b4165dfe3eabdb3f7ccb12b3f65b991371203","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 17:26:44.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b8f45d0762d07446c81777e919b1c94be52ae585","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-08-18 09:34:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5764f676e5aded16a7d3c566ff5d751e8aaeb586","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-08-18 09:34:42.000000000","message":"Patch Set 2:\n\nI just resolved the merge conflict","accounts_in_message":[],"_revision_number":2},{"id":"782c2d58e9c698d14d9f8b1b9bee721441f39b06","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-18 10:00:10.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/493865/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e4b4a5637d138e7f6890844d1108b12014a2085c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 10:53:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/2/check/gate-nova-docs-ubuntu-xenial/7e73a2a//doc/build/html/ : SUCCESS in 5m 12s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/2/check/gate-nova-pep8-ubuntu-xenial/3eee4d1/ : SUCCESS in 9m 02s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/2/check/gate-nova-python27-ubuntu-xenial/b64cfc3/ : SUCCESS in 14m 28s\n- gate-nova-python35 http://logs.openstack.org/65/493865/2/check/gate-nova-python35/62314a5/ : SUCCESS in 14m 43s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/2/check/gate-nova-tox-functional-ubuntu-xenial/86c972b/ : SUCCESS in 27m 30s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/18e3408/ : SUCCESS in 21m 34s","accounts_in_message":[],"_revision_number":2},{"id":"593645c8a3ab4fc79b4a48b80099ab29f1b0cd96","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-18 11:31:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/2/check/nova-out-of-tree-pvm/3f9cf53 : SUCCESS in 1h 56m 44s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/2/check/nova-in-tree-pvm/fb8beb8 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":2},{"id":"046a20b9de46c3624cf10ade8c6685366ad0a019","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-18 11:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/2/ : SUCCESS in 2h 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4719e406b727ba69f00bf788db06a14500e7087d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-18 11:54:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/65/493865/2/check-vote/ext-nova-zuul/11dd335 : FAILURE in 59m 29s","accounts_in_message":[],"_revision_number":2},{"id":"afb72ce24855079070019eeb34e10e60c1fce2db","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-18 12:00:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-18/493865/2/check/tempest-dsvm-full-nfv-xenial/1b314ad : SUCCESS in 2h 25m 40s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-18/493865/2/check/tempest-dsvm-intel-nfv-xenial/2e6db5d : SUCCESS in 47m 55s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-18/493865/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/ccf64b5 : SUCCESS in 1h 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce4d86dbe76ca6dde024682538197e8f3c747cef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-18 12:27:51.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"a909d0edaaf78f0664ef86423f899329cf4bf7a9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-18 12:45:17.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/bc88ec5 : FAILURE in 3h 10m 33s","accounts_in_message":[],"_revision_number":2},{"id":"78221ea9abcd1caa5fe739927b153181038036cb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-08-21 11:48:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : EXCEPTION\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"6e6ca38458d95fada404df08f99f2f6532a04e7c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 14:23:50.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"779580159230b426459be35590cd0296de37bafe","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-08-29 13:04:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52904b5a105b114c0efc6dc65e1a65e5bad6bccf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-08-29 13:05:40.000000000","message":"Patch Set 3:\n\njust resolved the merge conflict","accounts_in_message":[],"_revision_number":3},{"id":"61a3460aa2a67a209c50d0e3b76647b6710b78d4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 13:06:17.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/65e2f10 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"4f22cb6796ea29da20de8ef4995411267a55cabc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 13:35:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/3/check/gate-nova-docs-ubuntu-xenial/b33db02//doc/build/html/ : SUCCESS in 6m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/3/check/gate-nova-pep8-ubuntu-xenial/a7fc0a4/ : SUCCESS in 10m 34s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/3/check/gate-nova-python27-ubuntu-xenial/a8dd25a/ : SUCCESS in 15m 13s\n- gate-nova-python35 http://logs.openstack.org/65/493865/3/check/gate-nova-python35/03faba2/ : SUCCESS in 14m 28s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/3/check/gate-nova-tox-functional-ubuntu-xenial/ece4597/ : SUCCESS in 23m 04s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/3/check/gate-nova-tox-functional-py35-ubuntu-xenial/9bf4ba1/ : SUCCESS in 21m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f74b2a3f946c2fe1ae1766057d4d0870b3568ea5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-29 13:37:14.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"661d4586d89f9f311b273f38145965a66b7d84f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2017-08-29 13:42:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"27f029ca548fbe68f4393856445d1498f31180c3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-29 14:28:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/3/check/nova-out-of-tree-pvm/4541824 : SUCCESS in 1h 23m 12s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/3/check/nova-in-tree-pvm/c1ae887 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":3},{"id":"7791f400466a4bb716edf128460cd3f39d48e247","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-29 14:51:00.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/65/493865/3/check-vote/ext-nova-zuul/bc46a40 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":3},{"id":"d27cf584ea695bfb1beb7bad540647e1b79124cb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 15:06:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/3/ : SUCCESS in 2h 01m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e1736e72fe2f3aad20e09aa28ea8fb1d04e35256","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 16:18:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/493865/3/check/tempest-dsvm-full-nfv-xenial/0f034c8 : SUCCESS in 2h 21m 05s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/493865/3/check/tempest-dsvm-intel-nfv-xenial/b851473 : SUCCESS in 49m 51s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/493865/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/7f0b7e1 : SUCCESS in 1h 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8771cfde09243a5e613e6ff8b3c4fe040a48e73","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-29 16:44:59.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/493865/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"aface1b4564a6ac63ae83b42cb5132d82c071e20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-12 04:40:40.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"ab37464e5320d07b93b66b65d4099cd6104ec14c","author":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"date":"2017-09-12 06:15:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"19ddf33c96f3daf7c31875d74f14d336bac6aa73","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-09-12 16:44:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9fc36fd03d21ef13a91c2f41779b82947f330252","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-12 16:59:15.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/caa57ca : FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":4},{"id":"04bb8573bb9a4d5f6dd8f8d09eb819212b7a5ab6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-12 17:31:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/4/check/gate-nova-docs-ubuntu-xenial/9f3c451//doc/build/html/ : SUCCESS in 4m 51s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/4/check/gate-nova-pep8-ubuntu-xenial/961be61/ : SUCCESS in 9m 46s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/4/check/gate-nova-python27-ubuntu-xenial/13764c2/ : SUCCESS in 13m 17s\n- gate-nova-python35 http://logs.openstack.org/65/493865/4/check/gate-nova-python35/fb9ef2f/ : SUCCESS in 10m 24s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/4/check/gate-nova-tox-functional-ubuntu-xenial/9bad453/ : SUCCESS in 22m 32s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/4/check/gate-nova-tox-functional-py35-ubuntu-xenial/73256ff/ : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":4},{"id":"08a6f050d4b485576fd1f2758fb07ea0b93721d0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-12 17:31:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":4},{"id":"f682c34490a630954274b85b4a1efe3b32759cfb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-12 20:12:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/4/ : SUCCESS in 2h 25m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c1a42a2a4fd8fb01d291e2c8a27e21cf97518171","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-12 20:45:06.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/4/check/nova-out-of-tree-pvm/39e8897 : FAILURE in 3h 00m 45s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/4/check/nova-in-tree-pvm/8b7ead1 : FAILURE in 1h 05m 58s","accounts_in_message":[],"_revision_number":4},{"id":"ffc300dd2ee524f4ca0c18df1d9422a841b4d55b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-12 21:09:48.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"eeda4b3467977a32b4f62bc78db965afb89b2d88","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-15 07:51:03.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/493865/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"a0d60b7c20f8644cb645bb7baa59d15e5d360289","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2017-09-19 09:28:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"013e6820c0be9b2c2e4fb068dac92ae987857a3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 15:48:28.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"dff507d1b9ab5a0a935f1d6395a5e1afff8ff208","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-09-25 15:54:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b40b5d00c38ce4a7237e1a961ac7e8b3ed114476","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-09-25 15:54:43.000000000","message":"Patch Set 5:\n\nrebased and resolved merge conflict","accounts_in_message":[],"_revision_number":5},{"id":"35be40aa22a2f998d348b97e4f85770ece443efe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 16:22:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/5/check/gate-nova-docs-ubuntu-xenial/0d0169a//doc/build/html/ : SUCCESS in 5m 27s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/5/check/gate-nova-pep8-ubuntu-xenial/b5f7b41/ : SUCCESS in 10m 37s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/5/check/gate-nova-python27-ubuntu-xenial/b00b9d2/ : SUCCESS in 10m 42s\n- gate-nova-python35 http://logs.openstack.org/65/493865/5/check/gate-nova-python35/374c453/ : SUCCESS in 11m 42s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/5/check/gate-nova-tox-functional-ubuntu-xenial/73495ef/ : SUCCESS in 19m 09s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/5/check/gate-nova-tox-functional-py35-ubuntu-xenial/1fbfdf8/ : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":5},{"id":"a7e3a127e5c3011fba398d0a2e971f4ca4b8f1ec","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-25 16:23:03.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/493865/5/ : FAILURE in 28m 13s","accounts_in_message":[],"_revision_number":5},{"id":"500685f62153a31a470730d98235d9fb2cde0ee5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-25 17:27:27.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"90d3bc9a0b75a11558257f4d4b5caff5309aae22","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-25 17:48:12.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/caad7c3 : FAILURE in 1h 53m 24s","accounts_in_message":[],"_revision_number":5},{"id":"d7d4190dd36666e346329aa0795dca530e43aef1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 08:49:59.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"5a7cef8050d1970526bf400fa906d9048592a401","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-09-26 11:02:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cf215424ab92a091abf51a00595540d3feac1f76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 11:24:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/65/493865/6/check/gate-nova-docs-ubuntu-xenial/129f9b8//doc/build/html/ : SUCCESS in 4m 18s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/65/493865/6/check/gate-nova-pep8-ubuntu-xenial/f4b2427/ : SUCCESS in 8m 45s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/65/493865/6/check/gate-nova-python27-ubuntu-xenial/f6326e7/ : SUCCESS in 10m 42s\n- gate-nova-python35 http://logs.openstack.org/65/493865/6/check/gate-nova-python35/3d071ff/ : SUCCESS in 14m 08s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/65/493865/6/check/gate-nova-tox-functional-ubuntu-xenial/73a5e7b/ : SUCCESS in 19m 53s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/65/493865/6/check/gate-nova-tox-functional-py35-ubuntu-xenial/6e9996c/ : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":6},{"id":"d8f5bbca2a250db2569aca1ea0aa48a7e858b765","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-26 11:25:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":6},{"id":"84d90cd6c560451f599fe6a1f210de5a709037b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-26 11:26:10.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/65/493865/6/check-vote/ext-nova-zuul/2f9763b : FAILURE in 30s","accounts_in_message":[],"_revision_number":6},{"id":"d4c856bb98053628838bd672cec5587c0c6a85a2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-26 12:25:58.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"4f2e7a2f0e1612cd0ecc54a6e09c1425eb2489f4","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-26 13:12:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/6/ : SUCCESS in 2h 09m 20s","accounts_in_message":[],"_revision_number":6},{"id":"9887ab2158e56c2446cb3d40789d225c7f72b5e0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-26 13:28:29.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/493865/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1c5e026 : SUCCESS in 2h 23m 34s","accounts_in_message":[],"_revision_number":6},{"id":"768e6ac35e4707c67abb2b720c356c279fb3087d","author":{"_account_id":26490,"name":"Jason SUN","email":"sunjia@inspur.com","username":"sunjia"},"date":"2017-10-19 09:49:22.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bfd461fc2f30eacc3e8bad24bee3abdbde818706","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-19 15:07:50.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ac4dcc1f2a21671781a1c04dce1d2db4446726a2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 16:18:51.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/7/check/nova-out-of-tree-pvm/a75c7f5 : FAILURE in 40m 32s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/7/check/nova-in-tree-pvm/5465039 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":7},{"id":"1c45c777c29ccc1bae30af5dbffc62d958e4a560","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-19 16:40:51.000000000","message":"Patch Set 7: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7403 : FAILURE in 1h 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2e82f86b000a76cfa80bd365981d4a12a7e1c8aa","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 17:28:52.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/7/ : SUCCESS in 2h 20m 24s","accounts_in_message":[],"_revision_number":7},{"id":"0d35c3683e36c473d66c21467ffc58dc07b77560","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 18:33:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/7/check/openstack-tox-pep8/95d1f22/ : SUCCESS in 8m 05s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/7/check/openstack-tox-py27/02a0c92/ : SUCCESS in 10m 27s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/7/check/openstack-tox-py35/17b5e0a/ : SUCCESS in 14m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/7/check/build-openstack-sphinx-docs/75a4e84/html/ : SUCCESS in 5m 52s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/65/493865/7/check/legacy-tempest-dsvm-py35/3e5db71/ : SUCCESS in 1h 19m 12s\n- legacy-nova-tox-functional http://logs.openstack.org/65/493865/7/check/legacy-nova-tox-functional/32a5199/ : SUCCESS in 20m 44s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/65/493865/7/check/legacy-nova-tox-functional-py35/8b1ec66/ : SUCCESS in 22m 25s","accounts_in_message":[],"_revision_number":7},{"id":"38f1a3ccb0aca2e07e1400abd1df999b83b171cb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 18:34:14.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"d6eb4d191334f6c787f8be7aa1b06b7ec63ece46","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-19 19:38:39.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"e9c914061864170694475c6bb110a9cfb99d90af","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2017-11-25 05:38:32.000000000","message":"Patch Set 7:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":7},{"id":"457056f35e657490211a3ca37ca7578eab432690","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-25 08:08:34.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-9587 : SUCCESS in 2h 17m 40s","accounts_in_message":[],"_revision_number":7},{"id":"9417bc6b4b0a71d3622d23496ff5ab14dcf56f62","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-11-25 08:09:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/493865/7/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"f330ff59ff6ba0a9310c52d1ce3a266562a8430e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-11-27 16:43:55.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eefeed1b596e0c723328cb58218607ee5f3612a8","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-11-27 16:54:29.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nSome nits, but not at all critical.","accounts_in_message":[],"_revision_number":7},{"id":"e2cb10ae88c9c5a795e73c44d748af7669b87c54","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-12-04 12:47:35.000000000","message":"Patch Set 7: Code-Review+1\n\nlooks good","accounts_in_message":[],"_revision_number":7},{"id":"eeb9070824da78c9bf45f611d9f8277e6f126d02","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-08 03:09:46.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"11af33bddea6961e6796f18555c05bf04b4973a0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-30 16:56:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ba4d1e1f0eac5ca300d833241f543a097d1389b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-30 16:57:34.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5c46371235123a3c8f27d77bf5eddb24326f2582","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-30 16:57:56.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"0a27a78b0cd080431453498aa792873d9b4c622c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-30 18:12:35.000000000","message":"Patch Set 9:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/9/check/nova-out-of-tree-pvm/9ab3d32 : SUCCESS in 1h 02m 04s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/9/check/nova-in-tree-pvm/cd79e2a : FAILURE in 51m 58s","accounts_in_message":[],"_revision_number":9},{"id":"d4559a6ea0c5206f9bfb40e81eb2bd136a9f61e1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-30 18:13:38.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13319 : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":9},{"id":"6949c63ee99f85da034c3629b252c720d2222285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 19:06:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/9/check/openstack-tox-pep8/196fe54/ : SUCCESS in 10m 11s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/9/check/openstack-tox-py27/3e3f70e/ : SUCCESS in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/9/check/openstack-tox-py35/9f88fb1/ : SUCCESS in 17m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/9/check/build-openstack-sphinx-docs/6017135/html/ : SUCCESS in 7m 03s\n- tempest-full http://logs.openstack.org/65/493865/9/check/tempest-full/247c0c9/ : SUCCESS in 1h 14m 06s\n- neutron-grenade http://logs.openstack.org/65/493865/9/check/neutron-grenade/3c0ebb4/ : SUCCESS in 53m 06s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/65/493865/9/check/legacy-tempest-dsvm-py35/ff02ae1/ : SUCCESS in 1h 38m 46s\n- nova-tox-functional http://logs.openstack.org/65/493865/9/check/nova-tox-functional/ac0d57c/ : FAILURE in 20m 33s\n- nova-tox-functional-py35 http://logs.openstack.org/65/493865/9/check/nova-tox-functional-py35/fd51d80/ : FAILURE in 24m 27s","accounts_in_message":[],"_revision_number":9},{"id":"6d0d024f37273766fc98adb11031c8fab82e089d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-30 23:14:30.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/493865/9/ : SUCCESS in 3h 03m 59s","accounts_in_message":[],"_revision_number":9},{"id":"f15487cf051ab483525e3998bd48f22929547ddf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-01 12:37:48.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)\n\nthe failed test is relevant.","accounts_in_message":[],"_revision_number":9},{"id":"4b948621de404b27d8fba83bc3799d987ba81ca0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-07 14:47:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8c6675062b6f06a1518c7e34c779eb78b35e3fa9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-07 14:48:21.000000000","message":"Patch Set 10:\n\nrebased. Locally the tests are passing, let\u0027s see if it is the same in the CI","accounts_in_message":[],"_revision_number":10},{"id":"d61f472449b63e68851a8f98d9d81da13f6e49a3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-07 16:27:23.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14860 : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":10},{"id":"ee5d6b62a7fd251fa8ea223bfdad202be14eb79e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 16:29:04.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/10/check/openstack-tox-pep8/af6803e/ : SUCCESS in 8m 21s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/10/check/openstack-tox-py27/628e59a/ : SUCCESS in 13m 23s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/10/check/openstack-tox-py35/531a5f3/ : SUCCESS in 11m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/10/check/build-openstack-sphinx-docs/c0632ce/html/ : SUCCESS in 5m 53s\n- tempest-full http://logs.openstack.org/65/493865/10/check/tempest-full/64e88f0/ : SUCCESS in 1h 33m 57s\n- neutron-grenade http://logs.openstack.org/65/493865/10/check/neutron-grenade/cbd858e/ : SUCCESS in 41m 42s\n- tempest-full-py3 http://logs.openstack.org/65/493865/10/check/tempest-full-py3/d9ea94f/ : SUCCESS in 1h 13m 17s\n- nova-tox-functional http://logs.openstack.org/65/493865/10/check/nova-tox-functional/189b0fc/ : FAILURE in 21m 13s\n- nova-tox-functional-py35 http://logs.openstack.org/65/493865/10/check/nova-tox-functional-py35/34e34a6/ : FAILURE in 24m 56s","accounts_in_message":[],"_revision_number":10},{"id":"77b4262b8adc56787ff6249c8b619be6b5aaeb4d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-07 18:13:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-07/493865/10/check/tempest-dsvm-intel-nfv-xenial/240938a : SUCCESS in 29m 13s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/493865/10/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e971c52 : SUCCESS in 1h 04m 19s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/493865/10/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c03765c : SUCCESS in 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"491423a898ac746e6a44420756070deb274693a9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-03-07 21:54:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/10/check/nova-out-of-tree-pvm/af06ae9 : SUCCESS in 1h 27m 07s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/10/check/nova-in-tree-pvm/db699b1 : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":10},{"id":"3eb485a8bedc8ecfc6ec01f1c6b535010736c769","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-03-08 15:58:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4bb3e8375e63b37b2e6a1779b662a940b3b3bfb0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-03-08 16:54:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/11/check/nova-out-of-tree-pvm/5df090b : SUCCESS in 55m 18s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/11/check/nova-in-tree-pvm/0018685 : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":11},{"id":"b74e31dffc7b6d86f11ed402ff1aef233c868d48","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-08 17:14:45.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14940 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":11},{"id":"9638ece429139505169c651cdb6d6458ec26135e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-08 17:27:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/11/check/openstack-tox-pep8/1948f33/ : SUCCESS in 12m 52s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/11/check/openstack-tox-py27/98ea4e9/ : SUCCESS in 13m 14s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/11/check/openstack-tox-py35/87055e1/ : SUCCESS in 11m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/11/check/build-openstack-sphinx-docs/ab49916/html/ : SUCCESS in 7m 48s\n- tempest-full http://logs.openstack.org/65/493865/11/check/tempest-full/937b225/ : SUCCESS in 1h 17m 56s\n- neutron-grenade http://logs.openstack.org/65/493865/11/check/neutron-grenade/ce1774f/ : SUCCESS in 42m 24s\n- tempest-full-py3 http://logs.openstack.org/65/493865/11/check/tempest-full-py3/8cfc98b/ : SUCCESS in 1h 22m 07s\n- nova-tox-functional http://logs.openstack.org/65/493865/11/check/nova-tox-functional/28c4489/ : SUCCESS in 26m 35s\n- nova-tox-functional-py35 http://logs.openstack.org/65/493865/11/check/nova-tox-functional-py35/3b20c69/ : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":11},{"id":"abc6b3c04b9a3e1cb38b54f5495ddf5af6583450","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-08 18:37:08.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/65/493865/11/check-vote/ext-nova-zuul/6ba99ad : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":11},{"id":"d5477bfde7fdcf768d97c32456c278fe192a8a8c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-03-08 18:54:47.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-11 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":11},{"id":"8a97f8e9c54bd83e20026c1838aa080c0bcfa18a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-03-08 20:24:39.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/493865/11/ : FAILURE in 4h 12m 43s","accounts_in_message":[],"_revision_number":11},{"id":"bbda3de84d51e12a135914902af70cfc7cf0438f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-04-03 21:04:20.000000000","message":"Patch Set 11: Code-Review+1\n\nnice","accounts_in_message":[],"_revision_number":11},{"id":"226d07e0a9e43fbbbf414b724de36c674beb47d7","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-04-04 02:47:02.000000000","message":"Patch Set 11: Code-Review+1\n\nLooks good","accounts_in_message":[],"_revision_number":11},{"id":"f1c7df3af1923ea7a201b88b08567d3f6dd1e21a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-05-10 02:25:20.000000000","message":"Patch Set 11: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":11},{"id":"14e4feae79ac239f07e5e545f49f9523ad08b618","author":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"date":"2018-05-10 02:51:58.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"bc7544a5653619b11176d8bb21276d649374a4e4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-15 19:58:38.000000000","message":"Patch Set 11: Code-Review+2\n\n(3 comments)\n\nNice, gibi.  Clean and easy to follow.  Optional cleanups and a question inline.","accounts_in_message":[],"_revision_number":11},{"id":"2a2c568ee449ddda60acb9716bfdb8b151a3cf0b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-07-03 13:18:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"de04263be9a7f0fc1ea7dc2b87344d9d02891757","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-07-03 13:18:23.000000000","message":"Patch Set 11:\n\n(3 comments)\n\nresolved the merge conflict","accounts_in_message":[],"_revision_number":11},{"id":"aa76bfc08b4e57b5990ce844bd4cb9a440b94057","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-03 14:35:54.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/493865/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"bca37dcc7664aa7c5bc769402146c2f886ee105a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-03 14:57:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/12/check/openstack-tox-pep8/6dbc16d/ : SUCCESS in 10m 08s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/12/check/openstack-tox-py27/6622b6d/ : SUCCESS in 17m 16s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/12/check/openstack-tox-py35/1c7e1b7/ : SUCCESS in 12m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/12/check/build-openstack-sphinx-docs/d570899/html/ : SUCCESS in 8m 31s\n- tempest-full-py3 http://logs.openstack.org/65/493865/12/check/tempest-full-py3/ca79fb2/ : SUCCESS in 1h 16m 57s\n- nova-tox-functional http://logs.openstack.org/65/493865/12/check/nova-tox-functional/292ee90/ : SUCCESS in 16m 35s\n- nova-tox-functional-py35 http://logs.openstack.org/65/493865/12/check/nova-tox-functional-py35/ba822be/ : SUCCESS in 16m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/493865/12/check/openstack-tox-lower-constraints/4c686a7/ : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":12},{"id":"fcaae811c43a0fb51d38cc41bc36a4e3ce09f5c7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-03 15:26:26.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21055 : SUCCESS in 1h 39m 39s","accounts_in_message":[],"_revision_number":12},{"id":"fcb7b63d09bf88f43b075e81b6951a7d09caf83f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-03 15:48:45.000000000","message":"Patch Set 12:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-65-493865-12 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":12},{"id":"25ad7a2b2520eaa149aebba2daf3a555e54f15c9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-07-03 22:22:11.000000000","message":"Patch Set 12: Code-Review+2\n\nLovely","accounts_in_message":[],"_revision_number":12},{"id":"5823be0d8667bc250847c84ebce1319724a15cfd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-04 03:19:50.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/493865/12/ : FAILURE in 10h 05m 18s","accounts_in_message":[],"_revision_number":12},{"id":"2d4a71db5ab3807ffc27634cef8c6f261cf8581c","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-07-04 07:35:59.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"78865c181e64eee9994303bbb23f17a790f028b8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-07-04 10:05:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-07-03/493865/12/check/tempest-dsvm-intel-nfv-xenial/bf9c982 : SUCCESS in 31m 10s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-03/493865/12/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/7c2b344 : FAILURE in 1h 31m 05s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-03/493865/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/ef77d82 : SUCCESS in 57m 51s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6a5bc16be49b980aa9f08f72444034ed3af86b76","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-04 13:07:03.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/65/493865/12/check-vote/ext-nova-zuul/e1beaba : FAILURE in 19m 04s","accounts_in_message":[],"_revision_number":12},{"id":"1feca4fa0a88efec332100b443aabc60135cb415","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-07-05 00:23:08.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/493865/12/check/nova-out-of-tree-pvm/cd28d07 : SUCCESS in 1h 58m 20s\n- nova-in-tree-pvm http://184.172.12.213/65/493865/12/check/nova-in-tree-pvm/4f0b2f7 : SUCCESS in 1h 51m 07s","accounts_in_message":[],"_revision_number":12},{"id":"b1caa66935df00407a257027d313f9eec9ba5296","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-07-05 23:24:10.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nNicely done, looks correct to me","accounts_in_message":[],"_revision_number":12},{"id":"6f27ad42ad7bcc833783feb266c322bb544358ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-05 23:25:17.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"163bc67675074b272f5f190f765cf4ada3760f57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 00:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"7db89b8fbd62a6a907b4f45e8b600c9b06407f33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 00:55:09.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/493865/12/gate/openstack-tox-pep8/76d7aff/ : SUCCESS in 11m 19s\n- openstack-tox-py27 http://logs.openstack.org/65/493865/12/gate/openstack-tox-py27/c9af568/ : SUCCESS in 10m 24s\n- openstack-tox-py35 http://logs.openstack.org/65/493865/12/gate/openstack-tox-py35/4fc7bae/ : SUCCESS in 16m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/65/493865/12/gate/build-openstack-sphinx-docs/f8e7ccd/html/ : SUCCESS in 6m 20s\n- tempest-full-py3 http://logs.openstack.org/65/493865/12/gate/tempest-full-py3/be129d9/ : SUCCESS in 1h 14m 29s\n- nova-tox-functional http://logs.openstack.org/65/493865/12/gate/nova-tox-functional/729dd23/ : SUCCESS in 16m 48s\n- nova-tox-functional-py35 http://logs.openstack.org/65/493865/12/gate/nova-tox-functional-py35/117ced9/ : SUCCESS in 15m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/493865/12/gate/openstack-tox-lower-constraints/359bc15/ : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"63626a7bb9ffc84fabda224ec7f5b6fede2b4b64","revisions":{"b113525dfddf9df5bbb9149656d72cf42cd6d52e":{"kind":"REWORK","_number":1,"created":"2017-08-15 13:01:03.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/1"}}},"commit":{"parents":[{"commit":"3feed3a58a05b2a351a7f82d0b2848560af898e9","subject":"Merge \"Always use application/json accept header in report client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3feed3a58a05b2a351a7f82d0b2848560af898e9"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-15 12:08:51.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-15 12:08:51.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b113525dfddf9df5bbb9149656d72cf42cd6d52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b113525dfddf9df5bbb9149656d72cf42cd6d52e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3feed3a58a05b2a351a7f82d0b2848560af898e9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"0c7dd86367dce507e5c37ef28b1add0be3b467e0":{"kind":"REWORK","_number":2,"created":"2017-08-18 09:34:14.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/2"}}},"commit":{"parents":[{"commit":"08ec8a1ad3f3492b99db48d9e8fa132cb1bb3e8c","subject":"Merge \"Remove host filter for _cleanup_running_deleted_instances periodic task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08ec8a1ad3f3492b99db48d9e8fa132cb1bb3e8c"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-15 12:08:51.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-18 09:32:38.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c7dd86367dce507e5c37ef28b1add0be3b467e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c7dd86367dce507e5c37ef28b1add0be3b467e0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"08ec8a1ad3f3492b99db48d9e8fa132cb1bb3e8c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f898ecb3c6cca2353cd6369909e90f8ead792283":{"kind":"REWORK","_number":3,"created":"2017-08-29 13:04:57.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/3"}}},"commit":{"parents":[{"commit":"bd4a364e68398aae3dfe001bab4ebef31d086a9b","subject":"Merge \"Updates to scheduling workflow doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd4a364e68398aae3dfe001bab4ebef31d086a9b"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-29 13:02:19.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-29 13:02:19.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f898ecb3c6cca2353cd6369909e90f8ead792283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f898ecb3c6cca2353cd6369909e90f8ead792283"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bd4a364e68398aae3dfe001bab4ebef31d086a9b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3e8c221d56ea77223b2e349c183f6277a20c7b81":{"kind":"REWORK","_number":4,"created":"2017-09-12 16:44:33.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/4"}}},"commit":{"parents":[{"commit":"ea00db9ccad0989254d78090ea639333f66f001f","subject":"fake_notifier: Refactor wait_for_versioned_notification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea00db9ccad0989254d78090ea639333f66f001f"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-08-29 13:02:19.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-12 16:37:04.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e8c221d56ea77223b2e349c183f6277a20c7b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e8c221d56ea77223b2e349c183f6277a20c7b81"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ea00db9ccad0989254d78090ea639333f66f001f","is_merged_in_target_branch":true,"change_id":"Ie4676eed0039c927b35af7573f0b57fd762adbaa","change_number":489637,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"38472a1956c94e86eb844bb1f08b7ca28c8d2028":{"kind":"REWORK","_number":5,"created":"2017-09-25 15:54:21.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/5"}}},"commit":{"parents":[{"commit":"9d3307172210d3fadd05e39fdcbc47fe5c3e00bb","subject":"Merge \"Add functional migrate force_complete test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d3307172210d3fadd05e39fdcbc47fe5c3e00bb"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38472a1956c94e86eb844bb1f08b7ca28c8d2028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38472a1956c94e86eb844bb1f08b7ca28c8d2028"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d3307172210d3fadd05e39fdcbc47fe5c3e00bb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d2b2ccad9dfbd5d1fc85d25d5fee2f75087a368c":{"kind":"REWORK","_number":6,"created":"2017-09-26 11:02:38.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/6"}}},"commit":{"parents":[{"commit":"bb32286e9ea844e73722d06d57f015d17fb0e4fa","subject":"Merge \"Add functional for live migrate delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb32286e9ea844e73722d06d57f015d17fb0e4fa"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-26 11:02:38.000000000","tz":0},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2b2ccad9dfbd5d1fc85d25d5fee2f75087a368c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2b2ccad9dfbd5d1fc85d25d5fee2f75087a368c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bb32286e9ea844e73722d06d57f015d17fb0e4fa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1dc6d74c000b2c8c250e4869c9478b0c3d32aa86":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2017-10-19 15:07:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/7"}}},"commit":{"parents":[{"commit":"25afc31b202d66c294a8ac84895e2876cdd6681b","subject":"Merge \"Keep updating allocations for Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25afc31b202d66c294a8ac84895e2876cdd6681b"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-10-19 15:07:50.000000000","tz":0},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases works as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc6d74c000b2c8c250e4869c9478b0c3d32aa86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc6d74c000b2c8c250e4869c9478b0c3d32aa86"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"25afc31b202d66c294a8ac84895e2876cdd6681b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c7a5419a28fe8f0b4e32666fe7d75188cebb80b7":{"kind":"REWORK","_number":8,"created":"2018-01-30 16:56:59.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/8"}}},"commit":{"parents":[{"commit":"25afc31b202d66c294a8ac84895e2876cdd6681b","subject":"Merge \"Keep updating allocations for Ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25afc31b202d66c294a8ac84895e2876cdd6681b"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-30 16:56:22.000000000","tz":60},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases work as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c7a5419a28fe8f0b4e32666fe7d75188cebb80b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c7a5419a28fe8f0b4e32666fe7d75188cebb80b7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"25afc31b202d66c294a8ac84895e2876cdd6681b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1dc7eb27750fee4acc42693441ee93b438d0430f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-01-30 16:57:56.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/9"}}},"commit":{"parents":[{"commit":"91f7a999988b3f857d738d39984117e6c514cbec","subject":"Merge \"libvirt: don\u0027t attempt to live snapshot paused instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91f7a999988b3f857d738d39984117e6c514cbec"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-01-30 16:57:56.000000000","tz":0},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found mutliple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases work as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc7eb27750fee4acc42693441ee93b438d0430f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1dc7eb27750fee4acc42693441ee93b438d0430f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"91f7a999988b3f857d738d39984117e6c514cbec","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e1cc26e42e79e53df214e6379e7a78c922cbcde8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":10,"created":"2018-03-07 14:47:46.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/10"}}},"commit":{"parents":[{"commit":"130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a","subject":"Merge \"Move db MAX constants to own file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-07 14:47:25.000000000","tz":60},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found multiple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases work as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1cc26e42e79e53df214e6379e7a78c922cbcde8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1cc26e42e79e53df214e6379e7a78c922cbcde8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ea6086270eb265ff66ee25a59c0147563ec6d8cf":{"kind":"REWORK","_number":11,"created":"2018-03-08 15:58:35.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/11"}}},"commit":{"parents":[{"commit":"130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a","subject":"Merge \"Move db MAX constants to own file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-03-08 15:54:34.000000000","tz":60},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found multiple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases work as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea6086270eb265ff66ee25a59c0147563ec6d8cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea6086270eb265ff66ee25a59c0147563ec6d8cf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"130777ab0e6aa75e83f6cc33a5ab37fcdceb4e9a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"63626a7bb9ffc84fabda224ec7f5b6fede2b4b64":{"kind":"REWORK","_number":12,"created":"2018-07-03 13:18:07.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/65/493865/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/493865/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/493865/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/493865/12"}}},"commit":{"parents":[{"commit":"9382b7854286bba21235c4bfe39c585a6b894419","subject":"Merge \"Remove mox in unit/network/test_neutronv2.py (1)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9382b7854286bba21235c4bfe39c585a6b894419"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2017-09-25 15:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-07-03 13:17:10.000000000","tz":120},"subject":"cover migration cases with functional tests","message":"cover migration cases with functional tests\n\nAs we found multiple bugs in the server moving cases this patch adds\ntest to cover the migrate confirm and migrate revert cases as well.\nIt seems that these two cases work as expected.\n\nChange-Id: I40a244601e1612a87bd41cd6ad652202035a1ce4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63626a7bb9ffc84fabda224ec7f5b6fede2b4b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63626a7bb9ffc84fabda224ec7f5b6fede2b4b64"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9382b7854286bba21235c4bfe39c585a6b894419","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
