)]}'
{"id":"openstack%2Fnova~495516","triplet_id":"openstack%2Fnova~master~I69e95d07ba9a67cdd40649b0d1a2d789171d9984","project":"openstack/nova","branch":"master","topic":"bp/hurrah-for-privsep","hashtags":[],"change_id":"I69e95d07ba9a67cdd40649b0d1a2d789171d9984","subject":"Move lvm handling to privsep.","status":"MERGED","created":"2017-08-19 21:48:23.000000000","updated":"2017-10-20 16:44:25.000000000","submitted":"2017-10-20 16:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"495516-1508517865382-1be58a97","meta_rev_id":"f95f7535ab6dc4a055e90dd7aae5e71060e0b5ab","_number":495516,"virtual_id_number":495516,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"all":[{"value":0,"date":"2017-10-18 11:43:50.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-18 12:04:02.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-10-18 12:22:07.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-18 11:18:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2017-10-18 12:06:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2017-10-20 16:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-18 12:18:24.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-18 13:36:56.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2017-10-18 10:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2017-10-19 20:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2017-10-19 20:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-19 23:45:23.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-08-20 08:54:13.000000000","updated_by":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"reviewer":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"state":"REVIEWER"},{"updated":"2017-08-30 10:13:26.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-13 15:43:51.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2017-09-18 01:23:53.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-25 15:47:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2017-10-10 04:33:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-10 05:24:14.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 08:39:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 13:48:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 14:45:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 23:41:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 11:18:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 11:43:50.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 12:04:02.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 12:06:37.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-10-18 12:18:24.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 12:22:07.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-18 13:36:56.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-10-19 20:06:05.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2017-10-20 16:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d7995c5bef222c727bdb2968cbd0442f6efe03e","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-19 21:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92ec09b66f84c2364a3687e86da1087cefb3e724","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-19 22:39:42.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/495516/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"bd468fcdb8a0aeaa9d5947b68398d94922922f91","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-19 23:01:42.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/1/check/nova-out-of-tree-pvm/34ea7b8 : FAILURE in 1h 08m 24s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/1/check/nova-in-tree-pvm/b723c36 : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4c86826329aeeb465ba4eabf0ad46999ae01a002","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-19 23:06:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/1/check/tempest-dsvm-full-xenial/72c5de5/ : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bba45694b849a933a8d310aae455c23442019f81","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-19 23:29:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/1/ : SUCCESS in 1h 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5ad4460d99e7b0ee6d245c7499c44ff1cbb65ea0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-19 23:38:40.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/a0d9fc1 : SUCCESS in 1h 45m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9e0f2eaf004b3cbeafa36c73dbd66d40fd2cbeab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-19 23:45:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-docs-ubuntu-xenial/7b2aa2d/ : FAILURE in 41s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-pep8-ubuntu-xenial/4e3c09e/ : SUCCESS in 8m 35s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-python27-ubuntu-xenial/9643117/ : SUCCESS in 10m 34s\n- gate-nova-python35 http://logs.openstack.org/16/495516/1/check/gate-nova-python35/6311706/ : SUCCESS in 11m 48s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c2f16df/ : SUCCESS in 1h 22m 08s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/a6a3a56/ : SUCCESS in 51m 26s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/b52fd6b/ : SUCCESS in 1h 17m 48s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-tox-functional-ubuntu-xenial/8e11fa8/ : SUCCESS in 22m 54s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/c8339c7/ : SUCCESS in 18m 48s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/d6f2eda/ : SUCCESS in 58m 00s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/34b6fdd/ : FAILURE in 56m 35s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/2a13354/ : SUCCESS in 1h 12m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/a37d5ad/ : SUCCESS in 1h 13m 42s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/e06bc7f/ : SUCCESS in 1h 36m 37s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/8ddb4a5/ : SUCCESS in 1h 26m 43s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/7b7d8db/ : SUCCESS in 44m 30s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/9abf150/ : SUCCESS in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93395760410decf0c5bbf1a307688cb10f2f4b60","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-08-19 23:45:23.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-docs-ubuntu-xenial: https://bugs.launchpad.net/bugs/1384373\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"9246bc5e80795b3bc15e8253ee9be8f15515a0cb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-19 23:56:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/1/check/dsvm-tempest-neutron-network/2f15c26 : SUCCESS in 1h 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1a5fc7b2e58b4e739492e09f74cdb163b7cca311","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-20 00:11:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-full-nfv-xenial/a258d16 : SUCCESS in 2h 18m 48s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-intel-nfv-xenial/00bf9e8 : SUCCESS in 48m 18s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/83e6be3 : SUCCESS in 1h 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"320b9f1cfd029c66cb2e73590b40ad6ea5d2410e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-20 01:02:42.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/1/check/dsvm-tempest-xen/851f47c : FAILURE in 2h 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b1d3eb42465f1aa266a75fdfdc46af4212a1e542","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-20 02:10:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/1/check-nova/Nova-ML2-Sriov/199cfa0 : FAILURE in 42m 07s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/1/check-nova/Nova-MACVTAP-ML2-Sriov/548772a : SUCCESS in 52m 22s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"54d75c036773bf3ac6ea807524c12b0d68103c0d","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-20 05:38:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d5338dbe3ee6e32cb935e4efea66025c38455c8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-20 06:36:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/1/check-vote/ext-nova-zuul/daf3d35 : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"49e844dbdf3fe2cdfb74f7c310e300f7236ecd2a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-20 06:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/1/check/tempest-dsvm-full-xenial/f14f2cf/ : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3269296dd17e1dac2cdd3207a9022940a0cb8e69","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-20 07:01:54.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/1/check/nova-out-of-tree-pvm/a8243d2 : FAILURE in 1h 22m 46s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/1/check/nova-in-tree-pvm/4f3292d : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1b024956763d9519c6f6055cb63e4ea7604bdf8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-20 07:26:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/1/check/gate-nova-docs-ubuntu-xenial/ab76aec//doc/build/html/ : SUCCESS in 4m 36s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-pep8-ubuntu-xenial/c72f02c/ : SUCCESS in 8m 02s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-python27-ubuntu-xenial/7b5eef5/ : SUCCESS in 9m 58s\n- gate-nova-python35 http://logs.openstack.org/16/495516/1/check/gate-nova-python35/7996dfe/ : SUCCESS in 10m 54s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7061767/ : SUCCESS in 1h 25m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/ca180ba/ : SUCCESS in 47m 00s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/e5199ff/ : SUCCESS in 1h 15m 15s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-tox-functional-ubuntu-xenial/74424d3/ : SUCCESS in 22m 22s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/ccb9bd0/ : SUCCESS in 19m 13s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/b7d2770/ : SUCCESS in 58m 05s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/ade5a28/ : FAILURE in 55m 29s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/1c0c881/ : SUCCESS in 1h 10m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/0fa68dc/ : SUCCESS in 1h 08m 07s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/acbf0c1/ : SUCCESS in 1h 39m 01s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/1836cdf/ : SUCCESS in 1h 12m 42s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/0fd955a/ : SUCCESS in 39m 35s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/626fd6c/ : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d146bfed0d736e81ccedb96fc2e4dc634f7b0ed","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-20 07:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/1/ : SUCCESS in 1h 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0f8c58e84aa8e532e23d91bbc64ddb45f6bcbced","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-20 07:52:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-full-nfv-xenial/240d25c : SUCCESS in 2h 13m 09s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-intel-nfv-xenial/cc22270 : SUCCESS in 47m 42s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-19/495516/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/58f2d12 : SUCCESS in 1h 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8002eaf7cd92e26a3721c658786e78a7c17d3e47","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-20 08:21:45.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"160b8b9242f26923ef40cdf5d44825eb23cb6bc6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-20 08:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/1/check-vote/ext-nova-zuul/4db3615 : FAILURE in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d40725a04ca46e68de8aebeef4868e27047d10e7","author":{"_account_id":11869,"name":"zhubingbing","email":"zhu.bingbing@99cloud.net","username":"zhubingbing"},"date":"2017-08-20 08:54:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c3ab8df981d4f29fef5b7fb7f4c8dfb214c2c069","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-20 23:05:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fbb4768a6d0a1a621bdbe865b713748002dce9a0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-20 23:31:27.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/495516/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"abbb210e9a95791b92c77bb815a618a8fb4faa4b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-21 00:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/2/check/tempest-dsvm-full-xenial/9f1a5e1/ : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d7bf9f04bdef7ed0e2b51757eed0aaef36a961fd","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-21 00:49:26.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/2/check/nova-out-of-tree-pvm/368bfe9 : FAILURE in 1h 41m 52s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/2/check/nova-in-tree-pvm/4c36182 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":2},{"id":"684589cbdcc1a892cc1e613b5bc959af7cff9e3f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-21 00:54:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/2/check/gate-nova-docs-ubuntu-xenial/82cb4bb//doc/build/html/ : SUCCESS in 5m 58s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-nova-pep8-ubuntu-xenial/d74b890/ : SUCCESS in 10m 23s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-nova-python27-ubuntu-xenial/08c4e6a/ : SUCCESS in 11m 42s\n- gate-nova-python35 http://logs.openstack.org/16/495516/2/check/gate-nova-python35/8b53f07/ : SUCCESS in 11m 44s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c2a8ef0/ : SUCCESS in 1h 36m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/af50677/ : SUCCESS in 46m 55s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/c8c5327/ : SUCCESS in 1h 17m 35s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-nova-tox-functional-ubuntu-xenial/bed62e1/ : SUCCESS in 22m 34s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-nova-tox-functional-py35-ubuntu-xenial/f9f5c53/ : SUCCESS in 21m 00s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/780cef8/ : SUCCESS in 55m 38s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a6e0818/ : FAILURE in 58m 00s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-cells-ubuntu-xenial/61db237/ : SUCCESS in 1h 30m 05s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/aa6219e/ : SUCCESS in 1h 08m 55s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/d1272e7/ : SUCCESS in 1h 40m 56s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/eddd19a/ : SUCCESS in 1h 20m 41s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/def706a/ : SUCCESS in 38m 27s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/2/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/1d3e777/ : SUCCESS in 43m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"121d20cfa200996d30d85d3024c02213355c68e6","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-21 00:56:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/2/check-nova/Nova-ML2-Sriov/0209827 : SUCCESS in 1h 07m 19s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/2/check-nova/Nova-MACVTAP-ML2-Sriov/b364c80 : SUCCESS in 54m 03s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3f8644644793cf599b25d9a27f0e8dac5ec2d5e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-21 01:06:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/2/check/dsvm-tempest-neutron-network/9209456 : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ba4e6148a50f6451f639ef77da600158a31ab666","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-21 01:18:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/2/ : SUCCESS in 2h 11m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3f07fa23db2306d57be9b67acf2954470e8681a8","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-21 01:22:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/2/check/dsvm-tempest-xen/d6cc812 : SUCCESS in 2h 07m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b27a0f023b828ce01a7f2a38a6d89bd8ec16dcab","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-21 01:24:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-20/495516/2/check/tempest-dsvm-full-nfv-xenial/110e9b6 : SUCCESS in 2h 16m 50s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-20/495516/2/check/tempest-dsvm-intel-nfv-xenial/1d04264 : SUCCESS in 48m 38s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-20/495516/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/b2abd2b : SUCCESS in 1h 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d8867f76e991884e015cf5fcd7c06fdd70ce632","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-21 01:30:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/2/check-vote/ext-nova-zuul/97f4e52 : FAILURE in 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b6b3007adb779a6b4aea2d72ad6b99274c2e698f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-21 02:27:29.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"0559cdf81ac57a27e297264c88b57ec22e470c31","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-21 02:28:08.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9d08035 : FAILURE in 3h 18m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b27bfb65035a95fe590f004636f565c2b1c2b105","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-08-21 11:49:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : EXCEPTION\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"fba5a425b62690ca7f8a91ebfbfbf6a89e96795c","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-22 01:46:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"780080c1a16f2e36cfbb059c3c1629fa28cb7ce5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-22 03:02:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/3/check/tempest-dsvm-full-xenial/d4ded60/ : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":3},{"id":"05b636cf76a5e386bcd64362d053a4524e05c721","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-22 03:06:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/3/check-nova/Nova-ML2-Sriov/6b66225 : SUCCESS in 1h 01m 55s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/3/check-nova/Nova-MACVTAP-ML2-Sriov/7002716 : SUCCESS in 42m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"c91e7eb13af8d1833274fe47024c64afbfd3d7c4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 03:29:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/3/check/gate-nova-docs-ubuntu-xenial/7185469//doc/build/html/ : SUCCESS in 5m 11s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-nova-pep8-ubuntu-xenial/35182fb/ : SUCCESS in 8m 51s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-nova-python27-ubuntu-xenial/a2533f1/ : SUCCESS in 14m 06s\n- gate-nova-python35 http://logs.openstack.org/16/495516/3/check/gate-nova-python35/034da69/ : SUCCESS in 11m 42s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c061bc7/ : SUCCESS in 1h 35m 35s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/57f8dc0/ : SUCCESS in 51m 16s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/9d06922/ : SUCCESS in 1h 29m 08s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-nova-tox-functional-ubuntu-xenial/857010b/ : SUCCESS in 29m 01s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-nova-tox-functional-py35-ubuntu-xenial/21a2fde/ : SUCCESS in 20m 48s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/0160171/ : SUCCESS in 1h 00m 04s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/3/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/9079a82/ : FAILURE in 55m 50s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/f4bb392/ : SUCCESS in 1h 23m 22s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1a42e3d/ : SUCCESS in 1h 14m 10s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/4bf6ed0/ : SUCCESS in 1h 29m 19s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/ea00873/ : SUCCESS in 1h 20m 32s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/5e25dba/ : SUCCESS in 40m 59s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/c4b6b0a/ : SUCCESS in 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e40c58273e9b207c9db3793a7ae64f88c3153f9a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-22 03:52:01.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/3/check/nova-out-of-tree-pvm/f2903a6 : FAILURE in 1h 53m 34s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/3/check/nova-in-tree-pvm/1c21f9f : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":3},{"id":"208d4a7e0fba1b37e23ad96ab7bcae93c5cee6d8","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-22 03:56:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/3/check/dsvm-tempest-neutron-network/3f5d2db : SUCCESS in 2h 03m 30s","accounts_in_message":[],"_revision_number":3},{"id":"201f989cbf25116c645ca34d2905aa567386a4a9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-22 04:07:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/3/ : SUCCESS in 2h 14m 10s","accounts_in_message":[],"_revision_number":3},{"id":"ed75154640ab3dd33ba15c0a4c7f3b1998606f23","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-22 04:29:35.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/3/check/dsvm-tempest-xen/8cb464c : FAILURE in 2h 37m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d3ff6cb9211a462bd073e929670360e60ab674b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-08-22 04:36:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/3/check-vote/ext-nova-zuul/50b8233 : FAILURE in 1h 03m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3468d24d6e84b141d7ae1d1d4b4738950f34cf26","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-22 05:02:49.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"7fd2fcf6f516d88faca54ed909de6fc3b4e0635f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-22 05:20:13.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/4b82aae : FAILURE in 3h 30m 00s","accounts_in_message":[],"_revision_number":3},{"id":"01a8fb5cb63ff852eb7da6f8ec8ba5faec3a34df","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-22 06:14:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-21/495516/3/check/tempest-dsvm-full-nfv-xenial/bff1cb6 : SUCCESS in 2h 26m 01s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-21/495516/3/check/tempest-dsvm-intel-nfv-xenial/ae4c68a : SUCCESS in 50m 30s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-21/495516/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/48d4c2f : SUCCESS in 1h 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"234f80c708c70b52f33075fa17aedb2a7a221d76","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-22 21:59:20.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/495516/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bc2c3bbd56116a4e6a35b1157c503a42eb68d159","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-28 13:38:15.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6f72947e05360cb87aa0f28ce7314aeee4d4f517","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-28 20:52:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b92a607c548193f87caf06d1169ba12186d259df","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-28 20:52:43.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"a88351d3f8f7101bdbd1746863cf2a87291d616e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-28 20:52:47.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"51232777acec225541f4158cf58a0767af9d9567","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-28 20:52:51.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"43e12dfb8bcca355ce87b33978639f344978329c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-28 22:24:35.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/4/check/nova-out-of-tree-pvm/7b44e53 : FAILURE in 1h 30m 00s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/4/check/nova-in-tree-pvm/b700bea : FAILURE in 1h 15m 02s","accounts_in_message":[],"_revision_number":4},{"id":"07fc4c875a004d963783d05cd4e23758fc5c30e6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-28 22:44:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/4/check/dsvm-tempest-neutron-network/5b5373d : SUCCESS in 1h 49m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c5872716115b720ad1eb65256f59d6aa2928bb73","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-28 23:15:55.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/4/ : SUCCESS in 2h 21m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e83951bb18c5ff27c5bc65c8410b759bebbfd9de","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 00:24:12.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/dd1733a : FAILURE in 3h 29m 28s","accounts_in_message":[],"_revision_number":4},{"id":"ab9b0e3dc89605839658a7f6affbf8528e9b3f54","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-29 01:50:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/4/check-nova/Nova-ML2-Sriov/d8d2bd3 : SUCCESS in 1h 02m 39s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/4/check-nova/Nova-MACVTAP-ML2-Sriov/4b630c3 : SUCCESS in 56m 12s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"00f3482a3872684eea0d58b965bb1b622416f495","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 03:18:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/4/check/gate-nova-docs-ubuntu-xenial/a433027//doc/build/html/ : SUCCESS in 12m 37s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-nova-pep8-ubuntu-xenial/0c45cff/ : FAILURE in 9m 02s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-nova-python27-ubuntu-xenial/ded3dc8/ : FAILURE in 10m 13s\n- gate-nova-python35 http://logs.openstack.org/16/495516/4/check/gate-nova-python35/241f18e/ : FAILURE in 10m 35s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/670b01a/ : SUCCESS in 1h 27m 15s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3f8285c/ : FAILURE in 2h 51m 25s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/7d67958/ : SUCCESS in 1h 26m 46s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-nova-tox-functional-ubuntu-xenial/fa0aa15/ : SUCCESS in 30m 45s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-nova-tox-functional-py35-ubuntu-xenial/b528cd8/ : SUCCESS in 20m 37s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/14ae09f/ : SUCCESS in 58m 36s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/4/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/7f5163e/ : FAILURE in 59m 26s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-cells-ubuntu-xenial/e03b9b5/ : SUCCESS in 1h 22m 44s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/70f1011/ : SUCCESS in 1h 08m 45s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/313848e/ : SUCCESS in 1h 45m 26s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/caafcfa/ : SUCCESS in 1h 12m 24s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8bab272/ : SUCCESS in 41m 09s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/4/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/e04fc8e/ : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fc318baafd675cab1a88a2f80db487d1ae70f5d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-29 04:57:53.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/495516/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"0c69540cb2b5d45a5173ab85ffa69a2536080e8e","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-08-29 21:07:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d0fdbb7e3625530a72a89b40ae8aad407288d35c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:21.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"86de07f4136e0a995e768e2f3b07b0ec39f03d42","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:27.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"3afdb27a7d5d06d78ac473c286f6fefedfb615dc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:28.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"eea4f0c7ef4aa1f386a944d1c0cc1b3447ebd7b2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:46.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"491e106cc5e03e2deadabf18bcdb805a2b2b9d2f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 21:08:49.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"8982c9a0c19be223cdacf2d024786c56bb5aa1ec","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:51.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d10203353534cd6a559dae17ef5ddba5134fde0d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 21:08:53.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"462036c049840c965b140e8103d11332ffca156c","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-08-29 21:09:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"b1b45410b5fcc85f798271db82848c94b3c0a325","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-29 21:33:21.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/495516/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d23877b2ea34a5fd63c7f30dbce5737e73d505a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 22:37:31.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"5d202fa46694701c54348ceaa647a9a7d7c367cc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 10:13:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/495516/5/check/tempest-dsvm-full-nfv-xenial/2a8851c : SUCCESS in 2h 18m 31s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/495516/5/check/tempest-dsvm-intel-nfv-xenial/a63a1a0 : SUCCESS in 48m 22s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/495516/5/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/c8438e3 : SUCCESS in 1h 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"049c0fd7a0ed3531e471eb1a887be7f083a43830","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-31 00:27:36.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"ab6ac9b0425bcbb087b4a47d33c2e777a72a6977","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-03 07:44:43.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"79b732ddb5c4f97c5c543cf347c07c7fbd1911fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-03 07:45:05.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"4dee9bb5caa2a42b5bf4ddb7a42d91545c62d96d","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-03 07:45:13.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"c9c3d532b44080e4529f58cdec7adeb11fbfb728","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-03 07:45:23.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"f1cead42e931b6824f09d3c8eb657ff74d2ca179","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-03 07:45:51.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"659b7c120f8770e80a861346f0762abafdcf9e36","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-03 08:04:20.000000000","message":"Patch Set 6:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/69bc2bd : FAILURE in 17m 07s","accounts_in_message":[],"_revision_number":6},{"id":"13dcd18e5f9d4940db3ecbb9a68ad4e981da8def","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-03 08:53:46.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/495516/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"a11be875333ae9366effc5620b3060f1a3b4af78","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-03 09:45:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/6/ : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":6},{"id":"e45e213af5394c3053b3dad3c20a24f8fc87717b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-03 09:51:16.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/6/check/nova-out-of-tree-pvm/5c02983 : FAILURE in 2h 03m 44s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/6/check/nova-in-tree-pvm/4ab8f90 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":6},{"id":"4f7311f61c3b9936db29548593126f2af13ad12b","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-03 21:48:01.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"23819cdfe5f91b56d3febfa431bcae3e3a7b11bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-03 21:48:35.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"4b779f25d80afe2823ff034416e7519f0b7f8386","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-03 21:48:46.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"07e01b3aec338b798d263ca3560a236d9c99e78c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-03 21:49:02.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"340c6e8000b54e7e77b57f350904af135718ab31","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-03 22:10:21.000000000","message":"Patch Set 7:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/67e2a48 : FAILURE in 15m 40s","accounts_in_message":[],"_revision_number":7},{"id":"2bc84bf32968b584fd6f9c9702739a3c731ef644","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-03 22:57:07.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/495516/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"647aa89fd239f137a779cf5f04453841c425dd70","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-04 00:01:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/7/check/dsvm-tempest-neutron-network/f8c7008 : SUCCESS in 1h 56m 58s","accounts_in_message":[],"_revision_number":7},{"id":"c9e384b94243f60641d9dc2ca8fb556b8976d4d5","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-04 00:03:50.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/7/check/nova-out-of-tree-pvm/af1e66e : FAILURE in 2h 10m 50s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/7/check/nova-in-tree-pvm/b7da141 : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":7},{"id":"644db9c4e3515c3d9a508736bba8630cfcbea9ea","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-04 00:08:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/7/ : SUCCESS in 2h 16m 55s","accounts_in_message":[],"_revision_number":7},{"id":"55f041f761ba84b4b61fbc25b24498fe6f778416","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-04 02:27:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/7/check-nova/Nova-ML2-Sriov/e23029a : SUCCESS in 1h 06m 41s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/7/check-nova/Nova-MACVTAP-ML2-Sriov/750ddee : SUCCESS in 49m 47s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"79c5c3d1600f1a8b1fb288eaffd6c87024e3ff84","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-05 18:56:19.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"0c58399bb9c0ee91a3189c8d970556b944e09756","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-05 18:56:57.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"f9375c6f6398f6860a196d3faca7f3e6223fe7a7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-05 18:56:58.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/495516/8 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"b53b5567fb90c3bc3d94085d0e3be7703ec1909b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-05 18:57:06.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"3c3877680cc747933bfbc291a91c4e6baefd72fb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-05 18:57:28.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"f2ed1000235c157b4b5899e8404982360835aeea","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-05 18:57:34.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"53be3619f99241faa03df49a41ffd8de5ebc3614","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-05 21:08:17.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/8/check/nova-out-of-tree-pvm/21fa830 : FAILURE in 2h 06m 36s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/8/check/nova-in-tree-pvm/2cdfe18 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":8},{"id":"59350a45e8a874d523cd68d262afbb73024dda67","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-05 21:34:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/8/check/gate-nova-docs-ubuntu-xenial/176406f//doc/build/html/ : SUCCESS in 5m 25s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-nova-pep8-ubuntu-xenial/e33e50f/ : SUCCESS in 9m 43s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-nova-python27-ubuntu-xenial/dcf6a34/ : SUCCESS in 11m 59s\n- gate-nova-python35 http://logs.openstack.org/16/495516/8/check/gate-nova-python35/6726eb0/ : SUCCESS in 10m 35s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/6424cb0/ : SUCCESS in 1h 43m 44s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-grenade-dsvm-neutron-ubuntu-xenial/54e4845/ : SUCCESS in 57m 51s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-py35-ubuntu-xenial/392edc7/ : SUCCESS in 1h 19m 32s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-nova-tox-functional-ubuntu-xenial/da5f3f9/ : SUCCESS in 22m 32s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-nova-tox-functional-py35-ubuntu-xenial/cb7d7b9/ : SUCCESS in 20m 24s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/611ff4b/ : SUCCESS in 1h 04m 43s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/8/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/af35f95/ : FAILURE in 1h 06m 14s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-cells-ubuntu-xenial/2c6fb68/ : SUCCESS in 1h 20m 51s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/fbf4380/ : SUCCESS in 1h 25m 43s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/02c882d/ : SUCCESS in 1h 33m 05s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/8d02cec/ : SUCCESS in 1h 16m 48s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/341d3ed/ : SUCCESS in 37m 32s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/8/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/81fe9bb/ : SUCCESS in 45m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"070350deca9765a9488cccd99736135de9e68b81","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-06 00:09:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/8/check-nova/Nova-ML2-Sriov/d8afda7 : SUCCESS in 59m 46s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/8/check-nova/Nova-MACVTAP-ML2-Sriov/f88fc3f : SUCCESS in 53m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":8},{"id":"32d5c6c5545a89fb1e2a1cafdd90d86f5f8c0b9f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-06 01:40:41.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"ad822fe57788e737b6483d975d34feb891aa112b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-06 04:20:00.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/8/check-vote/ext-nova-zuul/18943d0 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":8},{"id":"0930421705adb516d4a19ce3b00e9d25db6ed4b6","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-06 17:03:50.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"1bc14eb5be9e61c495904cc3b6d4c648024d5c30","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 17:04:16.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"3328b52397c33d5f99c82cd4599fc93aafb85675","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-06 17:04:16.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"8e789fc4deae970f7eaf58d04882a7c499bc3a23","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-06 17:04:19.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"dc8d1d88c5cd310e621dd581b0a36fff812cfc25","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-06 17:04:57.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"def51e73b640caafd5627682919cf9b9cf3aa6ab","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-06 17:05:06.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"d4c63d15d3babb78a5b3fa64ed11636971eb7c34","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-06 17:05:58.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":9},{"id":"4c8088f4428b4c3c5ed94a6ac173490c2b7a6513","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-06 17:08:05.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"6da6c7498922a5d4cfb233485f9ce892b4ff999b","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-07 00:00:39.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"46c624ec847af2709c8fb211ea931ac55452d3e8","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-07 00:00:57.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"5a55fefb99ce36fa0a956d3da37f411aea524daa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-07 00:00:59.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"ed3f789a89560788ea008fdba3a5eef67c0a835e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-07 00:01:10.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"4038c528df9d0db60cbad6d8dd838356ce997226","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-07 00:01:51.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"fad8a3091f3eab05562f41441aeb960e901ccd3f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-07 00:01:55.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"e8bb66fa43af45c589eb627414d63d1320a71862","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-07 00:01:56.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"1bd6eb5d2ca5760b1f050e7b56e3f0ff51d84677","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-07 00:02:32.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"fcc00396c2fbf1586eb8f4b3d1a564e7606b1959","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-07 17:31:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6331c7c49864f476ece839060879f22306e432bf","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-07 17:31:57.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"f127de12eb24dd63d604b2bb2f589465ca821779","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-07 17:31:59.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"f6cba7008561aff53fc58442181ed0ea2112a97e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-07 17:32:07.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"039776c32db4f8558b327d27279f759aee92cf03","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-07 17:32:31.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"ec2dd17f98f608932d923bb255b20def99a142b9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-07 17:32:33.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"52c470d6e6bb074b3ed1cba8547cc03503ccd53b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-07 17:32:34.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"08875b5a7793c57393e0fd961aae5ce46e66b429","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-07 17:32:47.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"a73d66cf35e0b36d2250679f76bfe345115e51ff","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-07 17:33:10.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"67a30abaab4e1d03dd1825bbb44289086a816c72","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-08 03:11:11.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/495516/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"8f44967528c9ddd771ba72baacb4ebd87b9c7064","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-08 21:11:02.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"25069f1468345a0c80acc16244fdb0ad58333869","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-08 21:11:24.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"3fb2c6d82c912ce86497ba03db6bb5654b8b909a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-08 21:11:27.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"37d799c09882e7464530265d54181099ad8b1521","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-08 21:12:06.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"af3d9aa7d664194a199b1f4625e0353373fa77bb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-08 21:12:10.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"2dec41ef03d2b597ce6908f97133acf111c670da","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-08 21:12:13.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"e63bf73884f07f670408065918b023239423e19e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-08 21:12:34.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"9a96a4640875cc3c1b83a988f7f15f91c439b485","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-09 01:32:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/12/check/gate-nova-docs-ubuntu-xenial/4c1e3ab//doc/build/html/ : SUCCESS in 4m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-nova-pep8-ubuntu-xenial/d8fad60/ : SUCCESS in 11m 32s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-nova-python27-ubuntu-xenial/5dcc7a8/ : SUCCESS in 11m 37s\n- gate-nova-python35 http://logs.openstack.org/16/495516/12/check/gate-nova-python35/2aa8462/ : SUCCESS in 14m 22s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/1047274/ : SUCCESS in 1h 34m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-grenade-dsvm-neutron-ubuntu-xenial/bd460b6/ : SUCCESS in 56m 15s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-py35-ubuntu-xenial/3d858ca/ : SUCCESS in 1h 18m 19s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/12/check/gate-nova-releasenotes/e543f53//releasenotes/build/html/ : SUCCESS in 9m 31s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-nova-tox-functional-ubuntu-xenial/22f71d4/ : SUCCESS in 23m 47s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-nova-tox-functional-py35-ubuntu-xenial/a119c2e/ : SUCCESS in 25m 27s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/605093d/ : SUCCESS in 59m 56s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/12/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/6d3422f/ : FAILURE in 57m 36s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-cells-ubuntu-xenial/8701f1e/ : SUCCESS in 1h 19m 24s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/74def8b/ : SUCCESS in 1h 09m 43s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/6455698/ : SUCCESS in 1h 32m 10s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/e609874/ : SUCCESS in 1h 17m 23s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/fe72427/ : SUCCESS in 40m 02s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/12/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/d656452/ : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"952b333fee710734f9130608c1f961a2b0ceee66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-10 23:57:39.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"b9c16451453275277fe4fc1e9d6a996d2e6a94af","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-11 19:48:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5e54bd17eb02f79c1e89587854f32428599ba240","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-11 19:48:49.000000000","message":"Patch Set 13: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"64b155efd309ca6ebf66c112e7e968d704deb189","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-11 19:48:49.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"30ca843db9f327db6a2a2c5084c7116fe6bf684e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-11 19:48:59.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"b04b3da9e476f3672e1d2d3ab9ac1dca363dde5a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-11 19:51:00.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"d56bb97d16a5e5771c4500b19325e28a53557249","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-11 20:36:11.000000000","message":"Patch Set 13:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/edb1657 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":13},{"id":"7625a1ea3a113712b6addf0fa1e6f563d9941f90","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-11 22:29:19.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/13/ : SUCCESS in 2h 39m 41s","accounts_in_message":[],"_revision_number":13},{"id":"e5433c3617949ef7a1ce3f2235a919c6dd2dd125","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-11 22:37:55.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/13/check/dsvm-tempest-neutron-network/9118252 : SUCCESS in 2h 46m 08s","accounts_in_message":[],"_revision_number":13},{"id":"f677c85f31db4b03f432cb53d00bbf50a04dd132","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2017-09-13 15:43:51.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"b5fd69d38c01d391c52c25d634232774659afdf9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-13 16:43:29.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/13/check/tempest-dsvm-full-xenial/e1ebae1/ : SUCCESS in 56m 01s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/13/check/grenade-dsvm-xenial/367d79e/ : SUCCESS in 43m 11s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fc8c9f2c55ad44100bee3b6fb350bfc9f990392c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-13 16:44:21.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":13},{"id":"4eb5c308cc24246507d4757dcb459d3543a6e9d0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-13 17:02:22.000000000","message":"Patch Set 13:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/13/check/nova-out-of-tree-pvm/9a3839f : FAILURE in 1h 17m 31s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/13/check/nova-in-tree-pvm/6447cb7 : FAILURE in 51m 56s","accounts_in_message":[],"_revision_number":13},{"id":"3347d85c14d0255cbd78b020c0f54b3bf5f2dfd2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-13 17:52:02.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/13/ : SUCCESS in 2h 06m 53s","accounts_in_message":[],"_revision_number":13},{"id":"208a904d5d303bc134912845fc139eaa3ae6e45f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-13 17:56:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/13/check/gate-nova-docs-ubuntu-xenial/8b0d5e6//doc/build/html/ : SUCCESS in 6m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-nova-pep8-ubuntu-xenial/ec1a4f5/ : SUCCESS in 8m 54s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-nova-python27-ubuntu-xenial/5dbdb10/ : SUCCESS in 11m 35s\n- gate-nova-python35 http://logs.openstack.org/16/495516/13/check/gate-nova-python35/82fc572/ : SUCCESS in 13m 15s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/bc3f883/ : SUCCESS in 2h 04m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-grenade-dsvm-neutron-ubuntu-xenial/8901ca2/ : SUCCESS in 1h 00m 48s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-py35-ubuntu-xenial/c3867e5/ : SUCCESS in 1h 23m 09s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/13/check/gate-nova-releasenotes/0e6a397//releasenotes/build/html/ : SUCCESS in 10m 07s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-nova-tox-functional-ubuntu-xenial/ad5b35b/ : SUCCESS in 22m 18s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-nova-tox-functional-py35-ubuntu-xenial/a5bd6fa/ : SUCCESS in 20m 46s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/1dfd964/ : SUCCESS in 1h 06m 32s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/13/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/aba58de/ : FAILURE in 1h 01m 43s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-cells-ubuntu-xenial/ee4cc5d/ : SUCCESS in 1h 28m 41s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7c12299/ : SUCCESS in 1h 20m 59s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/57ab5be/ : SUCCESS in 1h 34m 00s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/c566505/ : SUCCESS in 1h 34m 18s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/a1089d9/ : SUCCESS in 41m 53s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/13/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/7078493/ : SUCCESS in 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c2800398f2ca5204f4c9ba82107f82812dd84bbb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-13 17:57:45.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":13},{"id":"7bb1491cd41f9c60a66b3810c4899a20d641c70e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-13 19:03:09.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"fd1902924550776b938dc11fcbdea26e7839e176","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-14 07:56:52.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/495516/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"2ea9e00baf0c0e5ab9e0c617eb5cfbbff432b76e","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-14 20:05:40.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b14c7792cd3a25478eb788ab74544325807d44b3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-14 20:09:13.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"f5ec3bf022443d8d7896b0026562c3a84184796f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-14 20:10:57.000000000","message":"Patch Set 14: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"96ec829a1f55e6519c886f4df1addc307cf518e2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-14 20:16:56.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"a288f7747aa9821f6a83a9165c08076df7988d5b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-14 20:17:20.000000000","message":"Patch Set 14:\n\nMerge failed. To try again use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":14},{"id":"1ebefc75ffaab7e26b5e8e54c3db7345edd5ee01","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-14 22:55:00.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"4df6d46ff2dc71ad3748d184eab7f0d1aaa6c038","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-14 23:43:17.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/15/check/tempest-dsvm-full-xenial/e35a2eb/ : FAILURE in 21m 16s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/15/check/grenade-dsvm-xenial/ac2bb30/ : FAILURE in 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"717a4468abd07c4da5db2ad9848e8b6293460c61","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-14 23:53:01.000000000","message":"Patch Set 15:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/15/check/nova-out-of-tree-pvm/378f2e8 : FAILURE in 29m 01s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/15/check/nova-in-tree-pvm/766d7f1 : FAILURE in 54m 02s","accounts_in_message":[],"_revision_number":15},{"id":"a5db25c096f99f9b6c97493941977e9b1a98076d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-15 00:14:18.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/15/check/gate-nova-docs-ubuntu-xenial/1bdeb6b//doc/build/html/ : SUCCESS in 4m 30s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-nova-pep8-ubuntu-xenial/34eba91/ : FAILURE in 9m 38s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-nova-python27-ubuntu-xenial/23b9b85/ : FAILURE in 4m 18s\n- gate-nova-python35 http://logs.openstack.org/16/495516/15/check/gate-nova-python35/3df3175/ : FAILURE in 4m 39s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ad0d14d/ : FAILURE in 23m 55s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-grenade-dsvm-neutron-ubuntu-xenial/d9963c9/ : FAILURE in 41m 54s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-py35-ubuntu-xenial/b404b42/ : FAILURE in 21m 54s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/15/check/gate-nova-releasenotes/f30577d//releasenotes/build/html/ : SUCCESS in 6m 24s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-nova-tox-functional-ubuntu-xenial/bef2819/ : FAILURE in 4m 10s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-nova-tox-functional-py35-ubuntu-xenial/5f3db08/ : FAILURE in 5m 41s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/74c42c2/ : FAILURE in 56m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/15/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/25c4d8e/ : FAILURE in 59m 59s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-cells-ubuntu-xenial/8ae0e47/ : FAILURE in 23m 48s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/feb4a8b/ : FAILURE in 32m 07s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/0584f8b/ : FAILURE in 31m 14s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/19c3ac0/ : FAILURE in 24m 51s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/bc2f542/ : FAILURE in 30m 25s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/15/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/7e36863/ : SUCCESS in 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"90261126980de77fe615ab27433b6ee32a3d5db4","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-15 00:14:51.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/15/check/dsvm-tempest-xen/2fc9aae : FAILURE in 45m 53s","accounts_in_message":[],"_revision_number":15},{"id":"456e904c131e9c5cd57522a8c279483df5dbd2c2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-15 00:41:56.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/15/check/dsvm-tempest-neutron-network/f3ad3d9 : FAILURE in 1h 29m 10s","accounts_in_message":[],"_revision_number":15},{"id":"6f0beec5b9fbd0307cf28966c0e961d6b720e517","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-15 01:07:03.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/15/ : SUCCESS in 2h 07m 45s","accounts_in_message":[],"_revision_number":15},{"id":"b25e2df3a33fe3578502564d61245b94570e3002","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-15 02:33:38.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/15/check-nova/Nova-ML2-Sriov/ef44bed : FAILURE in 40m 59s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/15/check-nova/Nova-MACVTAP-ML2-Sriov/2e03f42 : FAILURE in 10m 17s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"d3a420f69a977dadcea49dda011331aa2fc55fa0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-15 11:07:11.000000000","message":"Patch Set 15:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2022c40 : FAILURE in 3h 00m 52s","accounts_in_message":[],"_revision_number":15},{"id":"9682ec657758b55c4994f3085c201dae7aa1dacc","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-15 15:44:58.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"319d79a7a978a1c900eef76ff28582cbe2f39672","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-15 16:32:37.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/16/check/tempest-dsvm-full-xenial/69bcddf/ : FAILURE in 20m 28s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/16/check/grenade-dsvm-xenial/77145e2/ : FAILURE in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7727beddc625c45a64c6f7752b06053ea874c5c1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-15 16:57:56.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/16/check/gate-nova-docs-ubuntu-xenial/c55fb21//doc/build/html/ : SUCCESS in 6m 39s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-nova-pep8-ubuntu-xenial/854bc12/ : SUCCESS in 9m 52s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-nova-python27-ubuntu-xenial/0c09436/ : FAILURE in 5m 14s\n- gate-nova-python35 http://logs.openstack.org/16/495516/16/check/gate-nova-python35/cf5162c/ : FAILURE in 5m 16s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/ddc4725/ : FAILURE in 27m 16s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-grenade-dsvm-neutron-ubuntu-xenial/36f40ec/ : FAILURE in 40m 52s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-py35-ubuntu-xenial/8b845cb/ : FAILURE in 23m 23s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/16/check/gate-nova-releasenotes/893bd06//releasenotes/build/html/ : SUCCESS in 8m 45s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-nova-tox-functional-ubuntu-xenial/13c7e24/ : FAILURE in 5m 10s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-nova-tox-functional-py35-ubuntu-xenial/c9da8a6/ : FAILURE in 5m 01s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/ff2465a/ : FAILURE in 54m 47s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/16/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/98fef1b/ : FAILURE in 54m 29s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-cells-ubuntu-xenial/c54ceb1/ : FAILURE in 22m 55s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/7f1a6b0/ : FAILURE in 23m 51s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/2e783fd/ : FAILURE in 36m 07s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/5e16f90/ : FAILURE in 27m 02s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/8fdffd2/ : FAILURE in 31m 03s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/16/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/7c794c7/ : SUCCESS in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f30ab07c531956b0e63fbeab2ff5a9af742fe934","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-15 17:22:21.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/16/check/dsvm-tempest-xen/5edac4c : FAILURE in 59m 03s","accounts_in_message":[],"_revision_number":16},{"id":"d5298fc185d334a578d271a1a117988706f16f10","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-15 17:26:51.000000000","message":"Patch Set 16:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/16/check/nova-out-of-tree-pvm/e1481d2 : FAILURE in 27m 46s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/16/check/nova-in-tree-pvm/099b70f : FAILURE in 1h 38m 03s","accounts_in_message":[],"_revision_number":16},{"id":"e11ab39d2818447d7dc63b7021d82ea418cacb06","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-15 17:37:29.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/16/check/dsvm-tempest-neutron-network/9989b60 : FAILURE in 1h 29m 59s","accounts_in_message":[],"_revision_number":16},{"id":"fc48aeeb55363835fd9fa0d63604bb1f77b6e395","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-15 18:12:13.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/16/ : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":16},{"id":"a4801568a193f3cc7467c172be23b2b8a740e085","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-15 19:38:56.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"759ab719a24fb132c13408f3d1bfe253befda608","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-15 20:28:50.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/17/check/tempest-dsvm-full-xenial/7b0ae77/ : FAILURE in 26m 54s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/17/check/grenade-dsvm-xenial/1323d49/ : FAILURE in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b211153ea51bcaf1aa825c520a5b721bc365f76e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-15 21:39:36.000000000","message":"Patch Set 17:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/17/check/nova-out-of-tree-pvm/12a59ca : FAILURE in 28m 02s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/17/check/nova-in-tree-pvm/c7cab69 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":17},{"id":"372a9949cdaa4d4a8da2790e2f390096442e0915","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-15 22:20:40.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/17/check/gate-nova-docs-ubuntu-xenial/81937a9//doc/build/html/ : SUCCESS in 6m 43s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-nova-pep8-ubuntu-xenial/53429cb/ : SUCCESS in 9m 49s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-nova-python27-ubuntu-xenial/77c93c2/ : FAILURE in 5m 13s\n- gate-nova-python35 http://logs.openstack.org/16/495516/17/check/gate-nova-python35/ee4c30b/ : FAILURE in 5m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a732b66/ : SUCCESS in 1h 22m 37s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-grenade-dsvm-neutron-ubuntu-xenial/c59b7cf/ : FAILURE in 47m 57s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-py35-ubuntu-xenial/8396d5b/ : FAILURE in 26m 54s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/17/check/gate-nova-releasenotes/1d79499//releasenotes/build/html/ : SUCCESS in 8m 48s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-nova-tox-functional-ubuntu-xenial/95ed5eb/ : FAILURE in 13m 03s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-nova-tox-functional-py35-ubuntu-xenial/632e3d0/ : FAILURE in 4m 16s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/41cbcce/ : FAILURE in 1h 02m 37s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/17/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2ffd207/ : FAILURE in 59m 40s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-cells-ubuntu-xenial/d2137bd/ : FAILURE in 24m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ccf6496/ : FAILURE in 30m 08s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3a3ebaa/ : FAILURE in 31m 09s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/c21699a/ : FAILURE in 28m 31s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6217ce1/ : FAILURE in 31m 46s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/17/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/9b2a50b/ : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4859315b55d5d525a2791f83ac9dd3e1d96df768","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-15 22:55:41.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/17/check/dsvm-tempest-neutron-network/cc7019e : FAILURE in 1h 25m 07s","accounts_in_message":[],"_revision_number":17},{"id":"42191b610aa68479f7b601a149b8ef90dbae6dfb","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-16 02:20:54.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/17/check/dsvm-tempest-xen/851fd6c : FAILURE in 48m 38s","accounts_in_message":[],"_revision_number":17},{"id":"6951288fa10d2b05a02fb5e9e847d7fd95a01ae2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-16 03:53:34.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/17/ : SUCCESS in 2h 18m 40s","accounts_in_message":[],"_revision_number":17},{"id":"81bd985d9b57deb0263d92b6ea44390df974ed76","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-16 07:45:30.000000000","message":"Patch Set 17:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e4c4194 : FAILURE in 3h 54m 53s","accounts_in_message":[],"_revision_number":17},{"id":"db366aea9cc3cb3ed3b311d15bb32297662c80a7","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-16 15:57:42.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/495516/17 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"735b860696faf8d0aae789067ca7629b1150c714","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-17 23:05:07.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"6b5290d3edb5707b7b40323e992d81d0cdcc47d6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-17 23:18:39.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/18/check/tempest-dsvm-full-xenial/faba30d/ : FAILURE in 37s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/18/check/grenade-dsvm-xenial/78c8592/ : FAILURE in 36s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"7fb56825fc82d41dcfb3065e55fe2291c33ef78e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-18 00:11:59.000000000","message":"Patch Set 18:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/18/check/nova-out-of-tree-pvm/d6fe0c6 : FAILURE in 26m 37s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/18/check/nova-in-tree-pvm/976c214 : FAILURE in 1h 02m 35s","accounts_in_message":[],"_revision_number":18},{"id":"2a31996aea03bf196c6848027bb193a0ab54fae0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-18 00:23:07.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/18/check/gate-nova-docs-ubuntu-xenial/a74632c//doc/build/html/ : SUCCESS in 4m 40s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-nova-pep8-ubuntu-xenial/8f320d6/ : SUCCESS in 7m 34s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-nova-python27-ubuntu-xenial/41b0964/ : FAILURE in 3m 52s\n- gate-nova-python35 http://logs.openstack.org/16/495516/18/check/gate-nova-python35/dba79d9/ : FAILURE in 3m 28s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f06fccd/ : FAILURE in 25m 59s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2ce9b47/ : FAILURE in 50m 01s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-py35-ubuntu-xenial/bd3233a/ : FAILURE in 27m 31s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/18/check/gate-nova-releasenotes/f810668//releasenotes/build/html/ : SUCCESS in 8m 32s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-nova-tox-functional-ubuntu-xenial/d934476/ : FAILURE in 4m 14s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-nova-tox-functional-py35-ubuntu-xenial/290ca15/ : FAILURE in 4m 30s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/d0fde77/ : FAILURE in 1h 00m 01s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/18/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/ad6bdac/ : FAILURE in 52m 44s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-cells-ubuntu-xenial/7b2e0b3/ : FAILURE in 21m 00s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/38e8f19/ : FAILURE in 24m 54s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/3c63004/ : FAILURE in 32m 09s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/6c867ce/ : FAILURE in 26m 51s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/9a75532/ : FAILURE in 28m 50s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/18/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/34b2e47/ : SUCCESS in 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4a1f8d5304e9693200e08e14a4d27e04c31e386f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-18 00:28:04.000000000","message":"Patch Set 18:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/18/check/dsvm-tempest-xen/f10cd60 : FAILURE in 45m 32s","accounts_in_message":[],"_revision_number":18},{"id":"d85f2b37a697f7e529e1c6d55fe2e5c7ccdfc2f0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-18 00:52:41.000000000","message":"Patch Set 18:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/18/check/dsvm-tempest-neutron-network/464d84f : FAILURE in 1h 28m 09s","accounts_in_message":[],"_revision_number":18},{"id":"e66584b7b0798bbf8dcdadf208accc08ad131c3b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-18 01:13:43.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/18/ : SUCCESS in 2h 04m 08s","accounts_in_message":[],"_revision_number":18},{"id":"150600d518ac2c8f5a08f5e415966486f77006a0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-18 01:23:53.000000000","message":"Patch Set 18:\n\n* pci-test http://52.27.155.124/pci/495516/18 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"994527bfb85d49168dfc8e4143775be6f671ad8f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-18 03:06:37.000000000","message":"Patch Set 18:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/18/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ee5fa65 : FAILURE in 3h 54m 07s","accounts_in_message":[],"_revision_number":18},{"id":"0ef34b17ce78bfe822f0a9c993b40df272523972","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-19 01:39:38.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"fcdd3f84a439ddceaf7d9c569d097773523d9d7b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-19 02:35:16.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/19/check/tempest-dsvm-full-xenial/a714573/ : FAILURE in 39m 25s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/19/check/grenade-dsvm-xenial/206ed86/ : FAILURE in 39m 16s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"13dc6e8b783f3921d2929ce3720025bd2c386b9e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-19 02:36:48.000000000","message":"Patch Set 19:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/19/check/nova-out-of-tree-pvm/32ec73d : FAILURE in 27m 56s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/19/check/nova-in-tree-pvm/c4d253a : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":19},{"id":"624f873a852ba28d2284cb56e0c8d8f0c51f5a6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-19 03:10:41.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/19/check/gate-nova-docs-ubuntu-xenial/fec7683//doc/build/html/ : SUCCESS in 5m 41s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-pep8-ubuntu-xenial/188a959/ : SUCCESS in 10m 46s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-python27-ubuntu-xenial/b1ee05b/ : SUCCESS in 11m 38s\n- gate-nova-python35 http://logs.openstack.org/16/495516/19/check/gate-nova-python35/615e4ce/ : SUCCESS in 12m 13s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/0a7dedc/ : FAILURE in 57m 58s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-ubuntu-xenial/49d2fb2/ : FAILURE in 53m 42s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-py35-ubuntu-xenial/8eeefdb/ : FAILURE in 46m 22s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/19/check/gate-nova-releasenotes/0b1b460//releasenotes/build/html/ : SUCCESS in 10m 12s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-tox-functional-ubuntu-xenial/7e39f93/ : FAILURE in 45m 23s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-tox-functional-py35-ubuntu-xenial/2083256/ : SUCCESS in 20m 57s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f1a85ab/ : FAILURE in 57m 13s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/13cc124/ : FAILURE in 56m 19s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-cells-ubuntu-xenial/bd42072/ : FAILURE in 38m 06s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/144ad46/ : SUCCESS in 1h 15m 14s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/d6bd2a0/ : FAILURE in 1h 05m 54s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/57e215d/ : FAILURE in 47m 57s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/56adcb1/ : FAILURE in 41m 55s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/dfb516f/ : SUCCESS in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5d56aca3cbcd29654f573adcffacb7c764df88a3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-19 03:59:20.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/19/ : SUCCESS in 2h 13m 50s","accounts_in_message":[],"_revision_number":19},{"id":"3a92d47d6bc7f75aea3e16020e618eb28a439cc3","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-19 04:33:16.000000000","message":"Patch Set 19:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/19/check/dsvm-tempest-xen/22b6e5c : FAILURE in 2h 02m 58s","accounts_in_message":[],"_revision_number":19},{"id":"e631b305b38f9dd63050161c56a1eee4c99580c4","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-19 04:36:38.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/19/check/dsvm-tempest-neutron-network/1c6d551 : SUCCESS in 2h 09m 30s","accounts_in_message":[],"_revision_number":19},{"id":"06454958b034a5e772c6fe75f1e07a63ecad5d49","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-19 06:41:45.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/19/check-nova/Nova-ML2-Sriov/718aa11 : FAILURE in 47m 18s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/19/check-nova/Nova-MACVTAP-ML2-Sriov/d1dc546 : FAILURE in 49m 18s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":19},{"id":"cf2dc1d66b2042f1a082af5933dde2c0063950cf","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-21 03:19:59.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"f5b937bde9e6b5709db9ccd3207a6a60b8fdad07","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-21 04:22:45.000000000","message":"Patch Set 19:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/19/check/nova-out-of-tree-pvm/6b8840e : FAILURE in 30m 09s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/19/check/nova-in-tree-pvm/83295bc : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":19},{"id":"94959a82a69795895da2720933815e1765758763","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 04:58:41.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/19/check/gate-nova-docs-ubuntu-xenial/421b092//doc/build/html/ : SUCCESS in 10m 38s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-pep8-ubuntu-xenial/a92cfd5/ : SUCCESS in 14m 11s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-python27-ubuntu-xenial/fb4ba73/ : SUCCESS in 18m 53s\n- gate-nova-python35 http://logs.openstack.org/16/495516/19/check/gate-nova-python35/1d06cfc/ : SUCCESS in 18m 39s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/fd95232/ : FAILURE in 59m 00s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-ubuntu-xenial/f5d8033/ : FAILURE in 55m 56s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-py35-ubuntu-xenial/18ef185/ : FAILURE in 1h 18m 52s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/19/check/gate-nova-releasenotes/9e34620//releasenotes/build/html/ : SUCCESS in 12m 04s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-tox-functional-ubuntu-xenial/b23954e/ : SUCCESS in 30m 21s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-nova-tox-functional-py35-ubuntu-xenial/d02c3f0/ : SUCCESS in 22m 53s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f8db16f/ : FAILURE in 1h 27m 09s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/19/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/d154ae8/ : FAILURE in 59m 20s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-cells-ubuntu-xenial/5db9a3d/ : FAILURE in 36m 46s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/110683a/ : SUCCESS in 1h 16m 59s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/432223e/ : FAILURE in 1h 07m 27s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/4b366ef/ : FAILURE in 52m 44s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/d92318d/ : FAILURE in 39m 25s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/19/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/21de408/ : SUCCESS in 52m 43s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f46ce6cfb681fd07c441744e932e44db0d88000b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-21 06:02:43.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/19/ : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":19},{"id":"a7f1c30e1395ff1f170b6390308593a280fadd53","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-21 06:46:50.000000000","message":"Patch Set 19:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/19/check-vote/ext-nova-zuul/2919a3d : SUCCESS in 1h 22m 11s","accounts_in_message":[],"_revision_number":19},{"id":"4f8d40d241a41e332d0baed95c152353574df916","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-21 13:43:40.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/19/check/tempest-dsvm-full-xenial/9e43e42/ : FAILURE in 39m 57s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/19/check/grenade-dsvm-xenial/57c4c55/ : FAILURE in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"3cb63c0022eda5f02ad83e8600a3a55ca13d3ce2","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-23 04:14:56.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"2f1814c6d0f0778753748bd4bacc7f599cb78cad","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-23 05:21:01.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/20/check/tempest-dsvm-full-xenial/5bedf93/ : SUCCESS in 1h 00m 53s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/20/check/grenade-dsvm-xenial/b20a056/ : SUCCESS in 40m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ab93a50f6a7172805e5fc40a69ecabceb1fe57be","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-23 05:21:33.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":20},{"id":"2af7a422c3ad6d8e5a5face727c63e7d07c9a158","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-23 06:05:11.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/20/check/gate-nova-docs-ubuntu-xenial/8251f1b//doc/build/html/ : SUCCESS in 10m 52s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-pep8-ubuntu-xenial/af171df/ : SUCCESS in 9m 59s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-python27-ubuntu-xenial/abffbcb/ : SUCCESS in 13m 08s\n- gate-nova-python35 http://logs.openstack.org/16/495516/20/check/gate-nova-python35/29b4a87/ : SUCCESS in 12m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7440475/ : SUCCESS in 1h 42m 05s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7611f49/ : SUCCESS in 47m 05s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-py35-ubuntu-xenial/6d61a2a/ : SUCCESS in 1h 45m 16s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/20/check/gate-nova-releasenotes/944aa7a//releasenotes/build/html/ : SUCCESS in 6m 54s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-tox-functional-ubuntu-xenial/eb76b81/ : SUCCESS in 17m 54s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-tox-functional-py35-ubuntu-xenial/5793fe8/ : FAILURE in 45m 21s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/30c2390/ : SUCCESS in 1h 01m 32s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/fa1a8c9/ : FAILURE in 1h 00m 07s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-cells-ubuntu-xenial/f262e59/ : SUCCESS in 1h 09m 25s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d46a757/ : FAILURE in 1h 03m 44s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/afb1c09/ : SUCCESS in 1h 43m 04s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/348d884/ : SUCCESS in 1h 11m 45s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/30ae6df/ : SUCCESS in 41m 15s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/5d1f313/ : SUCCESS in 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a6b708a684971e34b0f14fc21e02735fd68890ef","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-23 06:25:10.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/20/check/dsvm-tempest-neutron-network/2e69bd7 : SUCCESS in 1h 58m 20s","accounts_in_message":[],"_revision_number":20},{"id":"6c134919ca32e79810e9f6d24c99e9ef5a632f3f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-23 06:36:20.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/20/ : SUCCESS in 2h 18m 46s","accounts_in_message":[],"_revision_number":20},{"id":"41d6f428073667d28e108f24be766e82b3207b4f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-23 06:42:54.000000000","message":"Patch Set 20:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/20/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/960de34 : SUCCESS in 2h 23m 16s","accounts_in_message":[],"_revision_number":20},{"id":"b0d63aae3dff8422aac080508130224edbfd217f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-23 06:52:40.000000000","message":"Patch Set 20:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/20/check/dsvm-tempest-xen/b332b52 : FAILURE in 2h 24m 42s","accounts_in_message":[],"_revision_number":20},{"id":"c978e9865cee5f080860afb8c4a9f85840d3d846","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-25 15:47:27.000000000","message":"Patch Set 20: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"4608ab8f883c5fbb2396ba9d896cbcf9765f425b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-25 15:56:31.000000000","message":"Patch Set 20:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/20/check/tempest-dsvm-full-xenial/1afe764/ : FAILURE in 5m 27s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/20/check/grenade-dsvm-xenial/e11e058/ : FAILURE in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4f69a8f8a643a85efeeb457df16e74f2f9c2728c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-25 16:18:23.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/495516/20/ : FAILURE in 30m 17s","accounts_in_message":[],"_revision_number":20},{"id":"61a27994ca80a5505c9e574ea68e006b0697063e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 17:33:59.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/20/check/gate-nova-docs-ubuntu-xenial/8271ef6//doc/build/html/ : SUCCESS in 4m 35s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-pep8-ubuntu-xenial/cf13eff/ : SUCCESS in 9m 46s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-python27-ubuntu-xenial/6f43348/ : SUCCESS in 11m 43s\n- gate-nova-python35 http://logs.openstack.org/16/495516/20/check/gate-nova-python35/1da64ff/ : SUCCESS in 13m 46s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a8bf3c2/ : FAILURE in 1h 17m 54s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-ubuntu-xenial/92c3043/ : SUCCESS in 53m 05s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-py35-ubuntu-xenial/5ba18f5/ : SUCCESS in 1h 17m 53s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/20/check/gate-nova-releasenotes/3b28038//releasenotes/build/html/ : SUCCESS in 7m 37s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-tox-functional-ubuntu-xenial/21a17f1/ : SUCCESS in 21m 34s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-nova-tox-functional-py35-ubuntu-xenial/ead2b2b/ : SUCCESS in 21m 02s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/857db9e/ : SUCCESS in 1h 11m 18s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/20/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/998a3bc/ : FAILURE in 52m 47s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-cells-ubuntu-xenial/423e7f8/ : SUCCESS in 1h 10m 58s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/bcc5dd4/ : FAILURE in 56m 19s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/4cb9af3/ : SUCCESS in 1h 41m 23s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/35a6b27/ : SUCCESS in 1h 08m 59s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/997d014/ : SUCCESS in 45m 42s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/20/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/af19b79/ : SUCCESS in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a9c9d37a3dab56a9ca068f9018c89b21f3e30a8e","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-27 00:42:29.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"510c813be006639d2d6c9203a18bb10f69ed7571","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-27 01:42:40.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/21/check/tempest-dsvm-full-xenial/2a08a89/ : SUCCESS in 55m 26s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/21/check/grenade-dsvm-xenial/c7dc5bb/ : SUCCESS in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"a126cc83d6f265eb43b32b22eb5ec507b2da12a7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 01:43:19.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":21},{"id":"6dfa67d9ef163a0d0697bef0e67f0353e92b9d5c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-27 02:26:46.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/21/check/gate-nova-docs-ubuntu-xenial/671b6cd//doc/build/html/ : SUCCESS in 4m 48s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-nova-pep8-ubuntu-xenial/f193f69/ : SUCCESS in 8m 41s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-nova-python27-ubuntu-xenial/b2c9ea2/ : SUCCESS in 16m 13s\n- gate-nova-python35 http://logs.openstack.org/16/495516/21/check/gate-nova-python35/a5078ae/ : SUCCESS in 11m 47s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/7bac39a/ : SUCCESS in 1h 21m 01s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3db374c/ : SUCCESS in 57m 24s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-py35-ubuntu-xenial/0495650/ : SUCCESS in 1h 13m 10s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/21/check/gate-nova-releasenotes/0fdb0a5//releasenotes/build/html/ : SUCCESS in 7m 35s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-nova-tox-functional-ubuntu-xenial/195d104/ : SUCCESS in 23m 18s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-nova-tox-functional-py35-ubuntu-xenial/d7ebca6/ : SUCCESS in 18m 59s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a9eaa8a/ : SUCCESS in 1h 33m 00s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/21/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/0900e32/ : FAILURE in 54m 31s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-cells-ubuntu-xenial/d7ca24d/ : SUCCESS in 1h 12m 19s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/299b007/ : FAILURE in 1h 10m 16s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/862b6fb/ : FAILURE in 1h 19m 23s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/936556b/ : SUCCESS in 1h 14m 34s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/1d682de/ : SUCCESS in 40m 51s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/21/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/f8735d4/ : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ec6cb7e55e9ae413a24a1d7402d43afaf0cf70c7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 02:27:25.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":21},{"id":"39706c93fb88b0ff5644159d2970968c47fa4c57","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-27 02:28:54.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/21/check/nova-out-of-tree-pvm/997f877 : SUCCESS in 1h 44m 09s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/21/check/nova-in-tree-pvm/623a1f5 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":21},{"id":"a80ae5c84bd6b16b053b30ad69ecf1df26810cf7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-27 02:46:52.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/21/check/dsvm-tempest-neutron-network/694b1bb : SUCCESS in 1h 52m 57s","accounts_in_message":[],"_revision_number":21},{"id":"d8db712f7b366cb22679878374188de1d309cbe8","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-27 03:02:57.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/21/check/dsvm-tempest-xen/b332070 : SUCCESS in 2h 07m 00s","accounts_in_message":[],"_revision_number":21},{"id":"69ea4cade4d4fda605c5f992a81ecea99c8352c2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-27 03:29:18.000000000","message":"Patch Set 21:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-21 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":21},{"id":"5d999aec6ea15d7d9b8ecf4048bd0d5d1e6d1db4","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-27 03:33:45.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"ab96718ffdb849260f822ba81d28081ac512dd64","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-27 04:34:14.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/22/check/tempest-dsvm-full-xenial/cd1fd2f/ : SUCCESS in 55m 47s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/22/check/grenade-dsvm-xenial/2ac054e/ : SUCCESS in 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ec4b2b3e99f3b4ad9f8ed265d02c5104a94b7fb0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 04:35:05.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":22},{"id":"a5e83190c9fd07983ec86c8e617b2bc32b7df1a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-27 05:17:12.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/22/check/gate-nova-docs-ubuntu-xenial/c1c3da9//doc/build/html/ : SUCCESS in 4m 59s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-nova-pep8-ubuntu-xenial/c20a63a/ : SUCCESS in 9m 44s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-nova-python27-ubuntu-xenial/8c027ed/ : FAILURE in 11m 53s\n- gate-nova-python35 http://logs.openstack.org/16/495516/22/check/gate-nova-python35/f070b68/ : FAILURE in 10m 32s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e19ce56/ : SUCCESS in 1h 17m 39s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-grenade-dsvm-neutron-ubuntu-xenial/31e92c1/ : SUCCESS in 50m 01s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-py35-ubuntu-xenial/fd4402f/ : SUCCESS in 1h 14m 51s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/22/check/gate-nova-releasenotes/b02037b//releasenotes/build/html/ : SUCCESS in 7m 56s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-nova-tox-functional-ubuntu-xenial/ee2e939/ : SUCCESS in 23m 58s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-nova-tox-functional-py35-ubuntu-xenial/d8787bc/ : SUCCESS in 21m 36s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/a3cac57/ : SUCCESS in 58m 32s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/22/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/47032db/ : FAILURE in 58m 59s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-cells-ubuntu-xenial/3fb310f/ : SUCCESS in 1h 13m 11s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/329ee7d/ : FAILURE in 1h 05m 33s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/efdb257/ : SUCCESS in 1h 32m 59s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/15bc68a/ : SUCCESS in 1h 10m 47s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/e3d1a82/ : SUCCESS in 42m 54s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/22/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/29065a0/ : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"1f48622b19f01c63df5c5aab584cc8deff9fb77d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-27 05:24:27.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/22/check/nova-out-of-tree-pvm/9f98cdc : SUCCESS in 1h 48m 31s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/22/check/nova-in-tree-pvm/94abe7d : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":22},{"id":"c46911dc47f2c31a27403d0a40d215e9ab87687b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-27 05:41:19.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/22/check/dsvm-tempest-neutron-network/e554162 : SUCCESS in 1h 55m 46s","accounts_in_message":[],"_revision_number":22},{"id":"2642cae8ecc36ffa4d74048d0ec1e9dd6cd74ba7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-27 05:43:29.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/22/ : SUCCESS in 2h 06m 29s","accounts_in_message":[],"_revision_number":22},{"id":"a3fca682fb404ed0e44ad78421c10d415175ad9b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-27 06:25:42.000000000","message":"Patch Set 22:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/22/check/dsvm-tempest-xen/1f528e3 : FAILURE in 2h 37m 50s","accounts_in_message":[],"_revision_number":22},{"id":"be85417571695d372eb75f54d108175adddc5818","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 06:32:12.000000000","message":"Patch Set 22:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/22/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/14b9017 : FAILURE in 2h 51m 50s","accounts_in_message":[],"_revision_number":22},{"id":"39284eb50d01539d3bea5997af639d61b9111256","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-27 17:39:39.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/22/check-nova/Nova-ML2-Sriov/2dd089d : SUCCESS in 54m 35s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/22/check-nova/Nova-MACVTAP-ML2-Sriov/4ff67f6 : SUCCESS in 44m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"e89d8a403f8208c2f51b11b5b7af9e3d6b660cd7","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-09-27 21:38:29.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"8abc9d5af076480345627921372d4f06ca5132dd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-27 22:44:17.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/23/check/tempest-dsvm-full-xenial/2f17b63/ : SUCCESS in 1h 00m 31s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/23/check/grenade-dsvm-xenial/055926f/ : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f43c288b299b0e471e8a2b56617dfe0f7ee05c7f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 22:44:54.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":23},{"id":"840cea9f54674b2170d3f3a0ea0a16d2efd1693b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-27 22:50:07.000000000","message":"Patch Set 23:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/23/check/nova-out-of-tree-pvm/bf047ef : FAILURE in 48m 01s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/23/check/nova-in-tree-pvm/d928d89 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":23},{"id":"641f8e3c24ffced072bda39f6fa28773e406dd39","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-27 23:26:56.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/23/check/gate-nova-docs-ubuntu-xenial/6a31d0c//doc/build/html/ : SUCCESS in 5m 51s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-nova-pep8-ubuntu-xenial/40a59d6/ : SUCCESS in 11m 52s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-nova-python27-ubuntu-xenial/43ce02b/ : SUCCESS in 13m 24s\n- gate-nova-python35 http://logs.openstack.org/16/495516/23/check/gate-nova-python35/13909df/ : SUCCESS in 12m 05s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/e78fab5/ : SUCCESS in 1h 30m 28s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7e08248/ : SUCCESS in 57m 30s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-py35-ubuntu-xenial/a3a8af2/ : SUCCESS in 1h 11m 46s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/23/check/gate-nova-releasenotes/c313293//releasenotes/build/html/ : SUCCESS in 8m 25s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-nova-tox-functional-ubuntu-xenial/cface31/ : SUCCESS in 19m 55s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-nova-tox-functional-py35-ubuntu-xenial/711488a/ : SUCCESS in 17m 49s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/7d80439/ : SUCCESS in 1h 02m 00s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/23/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/a55e142/ : FAILURE in 1h 00m 55s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-cells-ubuntu-xenial/0a19180/ : SUCCESS in 1h 14m 14s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d1a67be/ : FAILURE in 1h 02m 59s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/f8ab4d9/ : SUCCESS in 1h 38m 14s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/1e3698c/ : SUCCESS in 1h 39m 30s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/24fded2/ : SUCCESS in 48m 05s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/23/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/a813a92/ : SUCCESS in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"072bddc6fe7d4b40091536d1a78adc8530642074","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-27 23:27:32.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":23},{"id":"a5c8d6b88ca8122227572dc07913e377453cfdf5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-27 23:42:46.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/23/check/dsvm-tempest-neutron-network/cd95ad5 : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":23},{"id":"8a147164ec75f9db6d88bd767472212a2924047b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-28 00:15:10.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/23/check/dsvm-tempest-xen/bcf77cc : SUCCESS in 2h 20m 23s","accounts_in_message":[],"_revision_number":23},{"id":"0b25c43ea776994c1169fd8dfef4db88b2b8b58e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-28 00:29:53.000000000","message":"Patch Set 23:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/23/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/cc9470c : FAILURE in 2h 46m 29s","accounts_in_message":[],"_revision_number":23},{"id":"f1d76581559795c655e7055304cbac850b96d04e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-28 01:11:57.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"f63064e8df24e314347b3132bc812a15bf70967d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-28 02:20:44.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/495516/23/ : FAILURE in 4h 38m 45s","accounts_in_message":[],"_revision_number":23},{"id":"dd3a0fdae2e9346f9dd6352f7018b191eec33266","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-28 08:36:15.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/23/check-vote/ext-nova-zuul/c4961fd : FAILURE in 1h 13m 35s","accounts_in_message":[],"_revision_number":23},{"id":"b3685b057e9cf6aa28de43573fd4708898b6fe58","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-28 12:30:21.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/23/check-nova/Nova-ML2-Sriov/c79e9b4 : SUCCESS in 1h 05m 15s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/23/check-nova/Nova-MACVTAP-ML2-Sriov/3ce2506 : SUCCESS in 56m 47s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":23},{"id":"4158fb1c5f6194667a9405283d7d950671e68833","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-09-29 09:53:49.000000000","message":"Patch Set 23:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":23},{"id":"d6d3dce43aeb6f22abcf2a2120ad4052d7070615","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-29 11:01:32.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/23/check/tempest-dsvm-full-xenial/6cb1111/ : SUCCESS in 58m 11s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/23/check/grenade-dsvm-xenial/9e907b0/ : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"589008b528f9f2a81ee79f28627d5dfcefa916ae","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-29 11:02:16.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":23},{"id":"0f585fc339ed4f62fdbbd36144a462d6dc3b8d46","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-29 11:29:22.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/23/check-vote/ext-nova-zuul/880601c : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":23},{"id":"af2affc535d8e126c3dfcd42c9b12d4149d274a5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-29 12:26:53.000000000","message":"Patch Set 23:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/23/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c7c5c3e : FAILURE in 2h 24m 09s","accounts_in_message":[],"_revision_number":23},{"id":"e93f71e12cb59e55215dc72bfe9deb55959ea4e6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-29 13:07:25.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/23/ : SUCCESS in 2h 14m 50s","accounts_in_message":[],"_revision_number":23},{"id":"4aaaa20a4b36ad7c6fc6194423d74217c8657a34","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-10-10 03:02:47.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"31a0787ebd461c9225a14cee2151c0c8cde1ffac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-10 04:06:17.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/24/check/tempest-dsvm-full-xenial/5fe962e/ : SUCCESS in 1h 02m 28s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/24/check/grenade-dsvm-xenial/ddd9ca9/ : SUCCESS in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"b9fabddb57a0750f0940fa4742ab3c8e0d389cac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 04:06:45.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":24},{"id":"e568b2db40d3e9121dc010add34d6efad8cd4121","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-10 04:33:34.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/24/check/gate-nova-docs-ubuntu-xenial/3989fda//doc/build/html/ : SUCCESS in 5m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-pep8-ubuntu-xenial/4b618e3/ : SUCCESS in 8m 28s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-python27-ubuntu-xenial/d7015ca/ : SUCCESS in 12m 59s\n- gate-nova-python35 http://logs.openstack.org/16/495516/24/check/gate-nova-python35/5bbeab8/ : SUCCESS in 14m 19s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/295cdf6/ : SUCCESS in 1h 26m 24s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4f07555/ : SUCCESS in 48m 00s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-py35-ubuntu-xenial/ee49d67/ : SUCCESS in 1h 15m 36s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/24/check/gate-nova-releasenotes/81a8dc3//releasenotes/build/html/ : SUCCESS in 7m 40s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-tox-functional-ubuntu-xenial/a584d26/ : SUCCESS in 19m 51s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-tox-functional-py35-ubuntu-xenial/5fa15d9/ : SUCCESS in 17m 42s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/2b89825/ : SUCCESS in 59m 48s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/24af41e/ : FAILURE in 58m 26s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-cells-ubuntu-xenial/db57b12/ : SUCCESS in 35m 57s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/1a6cde5/ : FAILURE in 16m 07s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/f1b9d1e/ : SUCCESS in 1h 24m 27s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/2135ec9/ : SUCCESS in 1h 13m 55s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/2c013da/ : SUCCESS in 37m 36s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/afe76dc/ : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"513c044700ae836516443fdedb88c2254084fc73","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 04:34:03.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":24},{"id":"12f50a7097a646ea725c38ab74fdf29f3709e35a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-10 04:58:17.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/24/check/dsvm-tempest-neutron-network/af8e8d0 : SUCCESS in 1h 52m 28s","accounts_in_message":[],"_revision_number":24},{"id":"e3c3e2bb3a339652f6c04984f10feddc2c35517c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-10 05:11:54.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/24/check/nova-out-of-tree-pvm/a18d4b0 : SUCCESS in 2h 08m 12s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/24/check/nova-in-tree-pvm/c35d47d : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":24},{"id":"0a9ba7e2adc429e50ded623dd80f6e924785aae7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-10 05:24:14.000000000","message":"Patch Set 24:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/16/495516/24/check/dsvm-tempest-xen/3c68408 : FAILURE in 2h 12m 23s","accounts_in_message":[],"_revision_number":24},{"id":"9c816719e7c913500c742d39efb7380abfd22db6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-10-10 05:40:19.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/24/check-vote/ext-nova-zuul/c9eb3df : FAILURE in 1h 05m 02s","accounts_in_message":[],"_revision_number":24},{"id":"8cada9a9c9255c64ce84c3da76b41cc91ec3944a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-10 08:39:31.000000000","message":"Patch Set 24:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-16-495516-24 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":24},{"id":"0ca536d0c4682b1cf3522b8d30070996dbcfa463","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 08:46:05.000000000","message":"Patch Set 24: Verified-1\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/24/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/63b4153 : FAILURE in 2h 29m 57s","accounts_in_message":[],"_revision_number":24},{"id":"3779045be349f4c0fa4798c9a01182193e739a6f","author":{"_account_id":11351,"name":"Andreas Maier","email":"maiera@de.ibm.com","username":"maiera"},"date":"2017-10-10 08:48:10.000000000","message":"Patch Set 24:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":24},{"id":"f31498ba89d6e7db9e690496434e15517517a416","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-10 09:52:10.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/24/check/tempest-dsvm-full-xenial/deb8b70/ : SUCCESS in 1h 02m 39s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/24/check/grenade-dsvm-xenial/79f07cc/ : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c1495ce340d4e41e8302fd54fcaf1edab864ed82","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 09:52:37.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":24},{"id":"63d7b5fe726e9db57ad24354209a80d5c4f1580f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-10 10:37:27.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/16/495516/24/check/gate-nova-docs-ubuntu-xenial/2a8ceb2//doc/build/html/ : SUCCESS in 8m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-pep8-ubuntu-xenial/39c866b/ : SUCCESS in 12m 12s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-python27-ubuntu-xenial/badc6f0/ : SUCCESS in 13m 33s\n- gate-nova-python35 http://logs.openstack.org/16/495516/24/check/gate-nova-python35/55829a5/ : SUCCESS in 14m 20s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/64cb2a9/ : SUCCESS in 1h 22m 14s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-ubuntu-xenial/1113328/ : SUCCESS in 1h 00m 08s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-py35-ubuntu-xenial/3c617fd/ : SUCCESS in 1h 19m 08s\n- gate-nova-releasenotes http://docs-draft.openstack.org/16/495516/24/check/gate-nova-releasenotes/49ccdb2//releasenotes/build/html/ : SUCCESS in 12m 26s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-tox-functional-ubuntu-xenial/18a400d/ : SUCCESS in 26m 00s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-nova-tox-functional-py35-ubuntu-xenial/5ab9564/ : SUCCESS in 24m 00s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/34ea3a4/ : SUCCESS in 1h 02m 46s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/16/495516/24/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/e199809/ : FAILURE in 53m 02s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-cells-ubuntu-xenial/7b7ef6d/ : SUCCESS in 40m 31s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/d358e65/ : FAILURE in 15m 54s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/a0bb171/ : SUCCESS in 1h 44m 29s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/8a38786/ : SUCCESS in 1h 11m 30s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/7c3720d/ : SUCCESS in 41m 16s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/16/495516/24/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/121731e/ : SUCCESS in 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"09799276cc6c9551f27768eedad80436ae5f7457","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 10:37:58.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":24},{"id":"fcc59ebae6e6393a9a80bbc795f2196e5424d646","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 13:47:51.000000000","message":"Patch Set 24: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/495516/24/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d264ce3 : SUCCESS in 2h 04m 10s","accounts_in_message":[],"_revision_number":24},{"id":"41f2e5f60a53d9212f3c70351651ba8b73feb087","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 13:48:17.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":24},{"id":"4c8527bd3c107457335495d0370c1bf494dd6cae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-10-10 14:45:01.000000000","message":"Patch Set 24:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/16/495516/24/check-vote/ext-nova-zuul/6112afc : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":24},{"id":"21b8353ac36b23017d6ab0a9922d3b475c90cbcb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-10 18:05:09.000000000","message":"Patch Set 24:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/495516/24/ : FAILURE in 3h 24m 57s","accounts_in_message":[],"_revision_number":24},{"id":"a867258cde25f628052e2d1520f37758560eb8dd","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-10 23:41:15.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/16/495516/24/check-nova/Nova-ML2-Sriov/6a1171b : SUCCESS in 1h 11m 10s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/16/495516/24/check-nova/Nova-MACVTAP-ML2-Sriov/9157732 : SUCCESS in 56m 21s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"c50e9c79c56eaeefbaf3f652c1382e24820afa03","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2017-10-18 10:22:21.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"7ea11ce1449b8f4db4214b13a05eca62672c24a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-18 11:18:09.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/25/check/tempest-dsvm-full-xenial/1f6ac91/ : SUCCESS in 54m 22s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/16/495516/25/check/grenade-dsvm-xenial/5a2e814/ : FAILURE in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9977b4983d84f65ee991aba66ec4ba01685e8348","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-18 11:43:50.000000000","message":"Patch Set 25:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/16/495516/25/check/nova-out-of-tree-pvm/091ae4f : FAILURE in 1h 08m 13s\n- nova-in-tree-pvm http://184.172.12.213/16/495516/25/check/nova-in-tree-pvm/775a5d5 : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":25},{"id":"a0538927c3de6b700e9145a301116bcf8af52a12","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-18 12:04:02.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/16/495516/25/check/check-dsvm-tempest-vz7-exe-minimal/06c28b4 : SUCCESS in 1h 05m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":25},{"id":"dad6705b1897a2fb86ea9784da60f7c330ef855e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-18 12:05:36.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/16/495516/25/check/openstack-tox-pep8/bc27ac7/ : SUCCESS in 9m 06s\n- openstack-tox-py27 http://logs.openstack.org/16/495516/25/check/openstack-tox-py27/08e7781/ : SUCCESS in 11m 27s\n- openstack-tox-py35 http://logs.openstack.org/16/495516/25/check/openstack-tox-py35/f496445/ : SUCCESS in 11m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/495516/25/check/build-openstack-sphinx-docs/f47a762/html/ : SUCCESS in 4m 55s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-py35/0dd812c/ : SUCCESS in 1h 16m 13s\n- build-openstack-releasenotes http://logs.openstack.org/16/495516/25/check/build-openstack-releasenotes/2aa3330/html/ : SUCCESS in 7m 46s\n- legacy-nova-tox-functional http://logs.openstack.org/16/495516/25/check/legacy-nova-tox-functional/b7cda9f/ : SUCCESS in 19m 32s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/16/495516/25/check/legacy-nova-tox-functional-py35/9354c95/ : SUCCESS in 20m 32s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/16/495516/25/check/legacy-grenade-dsvm-neutron-multinode/11419d3/ : SUCCESS in 1h 00m 29s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/16/495516/25/check/legacy-grenade-dsvm-neutron-multinode-live-migration/efaf379/ : FAILURE in 1h 02m 07s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-cells/0d3c1a5/ : SUCCESS in 40m 11s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/2c7814a/ : FAILURE in 1h 21m 20s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-neutron-multinode-full/dd0b8bf/ : SUCCESS in 1h 38m 19s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-neutron-nova-next-full/74c60a3/ : FAILURE in 1h 00m 58s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-multinode-live-migration/1814099/ : SUCCESS in 42m 25s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7f24717/ : SUCCESS in 45m 59s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/495516/25/check/legacy-tempest-dsvm-neutron-full/af73c20/ : SUCCESS in 1h 32m 32s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/16/495516/25/check/legacy-grenade-dsvm-neutron/f814340/ : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":25},{"id":"75f6c20b76ec9341491eb6df1a0e8e449c7af187","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-18 12:06:37.000000000","message":"Patch Set 25: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7321 : FAILURE in 1h 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"dec864e5c64fa348057a93574fae585de8380699","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-18 12:18:24.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/16/495516/25/check/dsvm-tempest-neutron-network/23180cd : SUCCESS in 1h 53m 37s","accounts_in_message":[],"_revision_number":25},{"id":"086abf566045ae6ce160252dd9eed7dd0e554535","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-18 12:22:07.000000000","message":"Patch Set 25: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/495516/25/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e7a52b7 : SUCCESS in 1h 56m 25s","accounts_in_message":[],"_revision_number":25},{"id":"486cb34de2f14e5c5db10d7da03c7311a285e987","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-18 13:36:56.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/495516/25/ : SUCCESS in 3h 12m 15s","accounts_in_message":[],"_revision_number":25},{"id":"bb830b6aae4b4919b91177a13ce0ca73720eddc9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-10-19 20:06:05.000000000","message":"Patch Set 25: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"a6ec3ef15bce62c15952b2047d6b6f2ff60be479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-20 15:15:13.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"b8dcd305a49c52119c49fb9383a561e2f154a13b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-20 16:44:25.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/16/495516/25/gate/openstack-tox-pep8/511f2a9/ : SUCCESS in 8m 13s\n- openstack-tox-py27 http://logs.openstack.org/16/495516/25/gate/openstack-tox-py27/00a2766/ : SUCCESS in 13m 01s\n- openstack-tox-py35 http://logs.openstack.org/16/495516/25/gate/openstack-tox-py35/fefd2f1/ : SUCCESS in 12m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/16/495516/25/gate/build-openstack-sphinx-docs/d634ac9/html/ : SUCCESS in 5m 38s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/16/495516/25/gate/legacy-tempest-dsvm-py35/8dfa9a6/ : SUCCESS in 1h 19m 55s\n- build-openstack-releasenotes http://logs.openstack.org/16/495516/25/gate/build-openstack-releasenotes/1940d9e/html/ : SUCCESS in 7m 43s\n- legacy-nova-tox-functional http://logs.openstack.org/16/495516/25/gate/legacy-nova-tox-functional/9c2d6d1/ : SUCCESS in 21m 55s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/16/495516/25/gate/legacy-nova-tox-functional-py35/b1d641f/ : SUCCESS in 22m 16s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/16/495516/25/gate/legacy-tempest-dsvm-cells/df4f0e0/ : SUCCESS in 39m 14s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/16/495516/25/gate/legacy-tempest-dsvm-multinode-live-migration/68a6c8f/ : SUCCESS in 46m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/16/495516/25/gate/legacy-tempest-dsvm-neutron-full/d27935d/ : SUCCESS in 1h 18m 34s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/16/495516/25/gate/legacy-grenade-dsvm-neutron/2cbcf58/ : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":25},{"id":"66d344c1dae98f3a458c2954fc40b5722197976c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-20 16:44:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"8dff381426cf47dcc35470049f8b833a6590d0e3","revisions":{"b71d4a991c809ae3fc161aee21caa9277043bc21":{"kind":"REWORK","_number":1,"created":"2017-08-19 21:48:23.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/1"}}},"commit":{"parents":[{"commit":"605e45838bdf7cab3336a1822737cdd0cf388233","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/605e45838bdf7cab3336a1822737cdd0cf388233"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 21:47:55.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b71d4a991c809ae3fc161aee21caa9277043bc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b71d4a991c809ae3fc161aee21caa9277043bc21"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"605e45838bdf7cab3336a1822737cdd0cf388233","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f70b7f16439a3ceb2abde63dc8de3444816092b5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2017-08-20 23:05:31.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/2"}}},"commit":{"parents":[{"commit":"1ef3d5d3e9a4ffa966e39b3c34b338b3fdc4955b","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ef3d5d3e9a4ffa966e39b3c34b338b3fdc4955b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-20 23:04:47.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f70b7f16439a3ceb2abde63dc8de3444816092b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f70b7f16439a3ceb2abde63dc8de3444816092b5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1ef3d5d3e9a4ffa966e39b3c34b338b3fdc4955b","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"93925de761897dd490d88aaf801bfba2801fd80b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-08-22 01:46:29.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/3"}}},"commit":{"parents":[{"commit":"8e8c53eba270c38cd3c506c3bbcde0acb9f537f2","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e8c53eba270c38cd3c506c3bbcde0acb9f537f2"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-21 22:45:01.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93925de761897dd490d88aaf801bfba2801fd80b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93925de761897dd490d88aaf801bfba2801fd80b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8e8c53eba270c38cd3c506c3bbcde0acb9f537f2","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bd1d04afe64c78bf6dc5226ad6f3b74a83062d49":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-08-28 20:52:11.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/4"}}},"commit":{"parents":[{"commit":"4506bc09846e56c7457996a76a0c8326a52088ba","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4506bc09846e56c7457996a76a0c8326a52088ba"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-28 20:50:47.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1d04afe64c78bf6dc5226ad6f3b74a83062d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1d04afe64c78bf6dc5226ad6f3b74a83062d49"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4506bc09846e56c7457996a76a0c8326a52088ba","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"15206934ee238a13e1f5b55b588d82cdfa0ff766":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2017-08-29 21:07:40.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/5"}}},"commit":{"parents":[{"commit":"2be45ed21b1ba5215c32d35f21a2254f7aa49f66","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2be45ed21b1ba5215c32d35f21a2254f7aa49f66"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-29 21:06:00.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15206934ee238a13e1f5b55b588d82cdfa0ff766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15206934ee238a13e1f5b55b588d82cdfa0ff766"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2be45ed21b1ba5215c32d35f21a2254f7aa49f66","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3cd66a78f5065796b4028814eaec69a3cb205c8e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-09-03 07:44:43.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/6"}}},"commit":{"parents":[{"commit":"c5ea6f09de041067c2c56cb10b7e5daba4055d62","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c5ea6f09de041067c2c56cb10b7e5daba4055d62"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-02 11:21:50.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cd66a78f5065796b4028814eaec69a3cb205c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cd66a78f5065796b4028814eaec69a3cb205c8e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c5ea6f09de041067c2c56cb10b7e5daba4055d62","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2dc4929bed99d3175c316aaf758cf29bafdb8d2d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2017-09-03 21:48:01.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/7"}}},"commit":{"parents":[{"commit":"39b3feac8acecf5a173d4b0d30ae0a9ff6128438","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/39b3feac8acecf5a173d4b0d30ae0a9ff6128438"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-03 11:09:38.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dc4929bed99d3175c316aaf758cf29bafdb8d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dc4929bed99d3175c316aaf758cf29bafdb8d2d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"39b3feac8acecf5a173d4b0d30ae0a9ff6128438","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"df4a6766b1539898e8b351342315fbcfde4c54ec":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2017-09-05 18:56:19.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/8"}}},"commit":{"parents":[{"commit":"2345e8678bdb8c0086ffe16d67620f09563556bc","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2345e8678bdb8c0086ffe16d67620f09563556bc"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-05 18:55:22.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df4a6766b1539898e8b351342315fbcfde4c54ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df4a6766b1539898e8b351342315fbcfde4c54ec"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2345e8678bdb8c0086ffe16d67620f09563556bc","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2bd45751da285b736c67d27e5c9645865f4eb091":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2017-09-06 17:03:50.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/9"}}},"commit":{"parents":[{"commit":"23f15d655f17215ce28d564cba25fb66b300380b","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23f15d655f17215ce28d564cba25fb66b300380b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-06 17:01:13.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bd45751da285b736c67d27e5c9645865f4eb091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bd45751da285b736c67d27e5c9645865f4eb091"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"23f15d655f17215ce28d564cba25fb66b300380b","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cd8f59312021d3a13ff1a61145e937ca84fd64f7":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2017-09-07 00:00:39.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/10"}}},"commit":{"parents":[{"commit":"c90138af828f04d42b5ba75bcf920adff6d031f6","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c90138af828f04d42b5ba75bcf920adff6d031f6"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-06 19:50:28.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd8f59312021d3a13ff1a61145e937ca84fd64f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd8f59312021d3a13ff1a61145e937ca84fd64f7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c90138af828f04d42b5ba75bcf920adff6d031f6","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"06f4e1216087e45a9a93c46552ab20e9ac1a6bc8":{"kind":"REWORK","_number":11,"created":"2017-09-07 17:31:39.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/11"}}},"commit":{"parents":[{"commit":"f8183b0b0eb135f8accf80b07b5a241ddd56b820","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8183b0b0eb135f8accf80b07b5a241ddd56b820"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-07 17:27:13.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06f4e1216087e45a9a93c46552ab20e9ac1a6bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06f4e1216087e45a9a93c46552ab20e9ac1a6bc8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f8183b0b0eb135f8accf80b07b5a241ddd56b820","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6441fa7596c260c0cab3433eb53b30448a213496":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2017-09-08 21:11:02.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/12"}}},"commit":{"parents":[{"commit":"84e13f3b373e232a709bd65fb49574da86f6b1f5","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84e13f3b373e232a709bd65fb49574da86f6b1f5"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-08 21:05:35.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6441fa7596c260c0cab3433eb53b30448a213496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6441fa7596c260c0cab3433eb53b30448a213496"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"84e13f3b373e232a709bd65fb49574da86f6b1f5","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e44ca50c619356f4df09d848d7ed945e247b6267":{"kind":"REWORK","_number":13,"created":"2017-09-11 19:48:07.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/13"}}},"commit":{"parents":[{"commit":"826ae1bbeea41ca1d1feac2661f3055acf5084ce","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/826ae1bbeea41ca1d1feac2661f3055acf5084ce"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-11 19:47:27.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e44ca50c619356f4df09d848d7ed945e247b6267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e44ca50c619356f4df09d848d7ed945e247b6267"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"826ae1bbeea41ca1d1feac2661f3055acf5084ce","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b0e9ace189b92452c814cccea1d8fcbb99046d34":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":14,"created":"2017-09-14 20:05:40.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/14"}}},"commit":{"parents":[{"commit":"1efe7c7f667ce8de1a533c959fad005a4e8d969b","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1efe7c7f667ce8de1a533c959fad005a4e8d969b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-14 20:04:44.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0e9ace189b92452c814cccea1d8fcbb99046d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0e9ace189b92452c814cccea1d8fcbb99046d34"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1efe7c7f667ce8de1a533c959fad005a4e8d969b","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":19,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a45c4574cab1279ef57c663e11e5ede74ea8c679":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2017-09-14 22:55:00.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/15"}}},"commit":{"parents":[{"commit":"f1a32e5c9dc341f9dc57618a13e9eefa1f629465","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f1a32e5c9dc341f9dc57618a13e9eefa1f629465"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-14 22:54:15.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a45c4574cab1279ef57c663e11e5ede74ea8c679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a45c4574cab1279ef57c663e11e5ede74ea8c679"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f1a32e5c9dc341f9dc57618a13e9eefa1f629465","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":20,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0714d97da854e3b2c7482207120216a84cd972a7":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2017-09-15 15:44:58.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/16"}}},"commit":{"parents":[{"commit":"a00e499db62ca436f5396970fe5ffea6572cc6ae","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a00e499db62ca436f5396970fe5ffea6572cc6ae"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-15 15:44:23.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0714d97da854e3b2c7482207120216a84cd972a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0714d97da854e3b2c7482207120216a84cd972a7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a00e499db62ca436f5396970fe5ffea6572cc6ae","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":21,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b02fbd275261dc57f9997a0c2e7d7e8d2312a241":{"kind":"NO_CHANGE","_number":17,"created":"2017-09-15 19:38:56.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/17"}}},"commit":{"parents":[{"commit":"702f3721776363e84cb0e82f0b95a8f22a56074f","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/702f3721776363e84cb0e82f0b95a8f22a56074f"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-15 19:35:00.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b02fbd275261dc57f9997a0c2e7d7e8d2312a241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b02fbd275261dc57f9997a0c2e7d7e8d2312a241"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"702f3721776363e84cb0e82f0b95a8f22a56074f","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":22,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1ff22c6d21d92b025f269e3115c07cfc92164fd8":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2017-09-17 23:05:07.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/18"}}},"commit":{"parents":[{"commit":"a03f210e3b4b5941a4f608a63dd9c6f0bdbd7e21","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a03f210e3b4b5941a4f608a63dd9c6f0bdbd7e21"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-08-19 11:29:49.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-17 23:04:11.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ff22c6d21d92b025f269e3115c07cfc92164fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ff22c6d21d92b025f269e3115c07cfc92164fd8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a03f210e3b4b5941a4f608a63dd9c6f0bdbd7e21","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":23,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9593f57fe209106bcc1494efc7ac61dbf98c9e92":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2017-09-19 01:39:38.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/19"}}},"commit":{"parents":[{"commit":"a462a1bb302ec1174f61511c4bcbcd955d2db39b","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a462a1bb302ec1174f61511c4bcbcd955d2db39b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-18 13:29:16.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-18 22:32:11.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9593f57fe209106bcc1494efc7ac61dbf98c9e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9593f57fe209106bcc1494efc7ac61dbf98c9e92"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a462a1bb302ec1174f61511c4bcbcd955d2db39b","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":24,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d3b53ccd50f6541b9bd52ca205899f44b6998d09":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2017-09-23 04:14:56.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/20"}}},"commit":{"parents":[{"commit":"851ba03a2d32f7dd733957ee3c0bcd6865f3cb61","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/851ba03a2d32f7dd733957ee3c0bcd6865f3cb61"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-18 13:29:16.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-22 23:48:41.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d3b53ccd50f6541b9bd52ca205899f44b6998d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d3b53ccd50f6541b9bd52ca205899f44b6998d09"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"851ba03a2d32f7dd733957ee3c0bcd6865f3cb61","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":25,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6d800cffa39b7df5557f9788018b2a07e1dad72f":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2017-09-27 00:42:29.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/21"}}},"commit":{"parents":[{"commit":"9079a1cfdc043f0a001f2def4355d957d543681c","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9079a1cfdc043f0a001f2def4355d957d543681c"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d800cffa39b7df5557f9788018b2a07e1dad72f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d800cffa39b7df5557f9788018b2a07e1dad72f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9079a1cfdc043f0a001f2def4355d957d543681c","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":26,"change_status":"MERGED"}],"branch":"refs/heads/master"},"695f296f846863ba748418951f3c152910814476":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2017-09-27 03:33:45.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/22"}}},"commit":{"parents":[{"commit":"12b0f12b2989f2ec38d94c4f97637eca9260f33b","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12b0f12b2989f2ec38d94c4f97637eca9260f33b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-27 03:21:36.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/695f296f846863ba748418951f3c152910814476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/695f296f846863ba748418951f3c152910814476"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12b0f12b2989f2ec38d94c4f97637eca9260f33b","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":27,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6497cccd2a0656b00344ec25d0637bf88a502663":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2017-09-27 21:38:29.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/23"}}},"commit":{"parents":[{"commit":"d6f8bcfe829a7c28bc8d85f791c3ff374459ba38","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6f8bcfe829a7c28bc8d85f791c3ff374459ba38"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-27 21:30:01.000000000","tz":600},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6497cccd2a0656b00344ec25d0637bf88a502663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6497cccd2a0656b00344ec25d0637bf88a502663"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d6f8bcfe829a7c28bc8d85f791c3ff374459ba38","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":28,"change_status":"MERGED"}],"branch":"refs/heads/master"},"afd0c40fba9892f57e57304a9363de87be2bbc48":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2017-10-10 03:02:47.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/24"}}},"commit":{"parents":[{"commit":"6591e14a5b9251b520ba27bf7e784b237d3c0d3f","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6591e14a5b9251b520ba27bf7e784b237d3c0d3f"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-10-09 22:45:49.000000000","tz":660},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afd0c40fba9892f57e57304a9363de87be2bbc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afd0c40fba9892f57e57304a9363de87be2bbc48"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6591e14a5b9251b520ba27bf7e784b237d3c0d3f","is_merged_in_target_branch":false,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":29,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8dff381426cf47dcc35470049f8b833a6590d0e3":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2017-10-18 10:22:21.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/16/495516/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/495516/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/495516/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/495516/25"}}},"commit":{"parents":[{"commit":"7ad72b092201f87530674a245e2904c6401d567c","subject":"Cleanup mount / umount and associated rmdir calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ad72b092201f87530674a245e2904c6401d567c"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-09-26 20:53:54.000000000","tz":600},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2017-10-18 06:53:38.000000000","tz":660},"subject":"Move lvm handling to privsep.","message":"Move lvm handling to privsep.\n\nFollowing the same pattern as the others.\n\nChange-Id: I69e95d07ba9a67cdd40649b0d1a2d789171d9984\nblueprint: hurrah-for-privsep\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dff381426cf47dcc35470049f8b833a6590d0e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dff381426cf47dcc35470049f8b833a6590d0e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ad72b092201f87530674a245e2904c6401d567c","is_merged_in_target_branch":true,"change_id":"Ib5e585fa4bfb99617cd3ca983674114d323a3cce","change_number":494423,"patch_set_number":30,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
