)]}'
{"id":"openstack%2Fnova~498818","triplet_id":"openstack%2Fnova~master~Ic318e04be73639de7d26e53e1f0b152d9803fd2f","project":"openstack/nova","branch":"master","topic":"bp/doc-migration","hashtags":[],"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","subject":"doc: Add configuration index page","status":"MERGED","created":"2017-08-29 14:07:14.000000000","updated":"2018-01-04 00:38:43.000000000","submitted":"2017-09-06 19:08:30.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c1c77718a4640f871778d3f5111648b987f69a61","_number":498818,"virtual_id_number":498818,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-09-06 20:53:29.000000000","post_submit":true,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-09-06 19:08:30.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2017-09-07 18:03:32.000000000","post_submit":true,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":2,"date":"2017-09-06 18:17:10.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-09-06 13:41:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2017-09-06 13:41:24.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":1,"date":"2017-09-06 13:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2017-09-06 18:17:10.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-30 13:34:56.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-01 01:12:47.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2017-09-04 10:42:29.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-04 12:01:08.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-04 13:42:56.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-09-05 06:40:04.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-09-06 11:50:09.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2017-09-06 18:17:10.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2017-09-06 19:08:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-06 20:53:29.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-07 18:03:32.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3ac97184ccd912e476c8937bb293df1e9cc9bee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-29 14:07:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed22354452405b36dfc095e1e39a16265777ee93","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 14:08:06.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2ef1026 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"220e22b45014b4bf7af27e688b7eb186fb28b67d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 14:20:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/1/check/gate-nova-docs-ubuntu-xenial/318b2f5//doc/build/html/ : SUCCESS in 7m 39s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/1/check/gate-nova-pep8-ubuntu-xenial/3e0c638/ : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5123a08765fa758ab0a3b1c57b50d6ae1cc23343","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-29 14:21:49.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-498818-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"935757a3c6fda9de526a249077c9823c7239d669","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 16:13:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498818/1/ : SUCCESS in 2h 05m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ec98c09dce1eb0a644f1ebaf7464f45c6b3b86e5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-29 16:18:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7bf4449597baf6f494fb5cd79e270c70d2785e3c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 16:38:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://logs.openstack.org/18/498818/2/check/gate-nova-docs-ubuntu-xenial/c4cfaa9/ : FAILURE in 5m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/2/check/gate-nova-pep8-ubuntu-xenial/41a6855/ : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d2241d343c47b32da287f2775817dd1906e491fe","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-29 17:43:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/498818/2/check/nova-out-of-tree-pvm/53555d0 : SUCCESS in 1h 23m 26s\n- nova-in-tree-pvm http://184.172.12.213/18/498818/2/check/nova-in-tree-pvm/b91bbd4 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9e2c81ef94d3f8b47f960964645891f9d09561bc","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-29 18:40:44.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/498818/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c3d919ef9e3e16d14bdff8a247c66464cd90961b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 18:52:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498818/2/check/tempest-dsvm-full-nfv-xenial/e70ad83 : SUCCESS in 2h 19m 41s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498818/2/check/tempest-dsvm-intel-nfv-xenial/9c6711b : SUCCESS in 49m 11s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498818/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/0a4d8e8 : SUCCESS in 1h 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8073dfebbef4aeddc64d58b097016fa1e98f084","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 19:09:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498818/2/ : SUCCESS in 1h 50m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c02866146c95ee596ccc2674ec98fd0af7ca85cb","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 21:15:43.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ab1b56b : FAILURE in 3h 22m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c7efca4f4088d3f98fafc4208fa1841d440870ad","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-30 10:47:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2569df3ffbd0cb21673fee15d033320dbaa9d5ef","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 10:53:35.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f42b2d4 : FAILURE in 1m 56s","accounts_in_message":[],"_revision_number":3},{"id":"34496872deb4b13904d5d91bedab297a2407062e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 11:00:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/3/check/gate-nova-docs-ubuntu-xenial/ddbf327//doc/build/html/ : SUCCESS in 4m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/3/check/gate-nova-pep8-ubuntu-xenial/6c3691c/ : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":3},{"id":"0572189b5dab59037169dad07549a9dadee28c33","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-30 11:02:16.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-498818-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"07cc759c0ed8dc769a2c94dec389ee21c5354c60","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-08-30 11:22:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nnice to have this index in main index as most of this content and links similar to main index section \u0027Reference Material\u0027\n\n\nReference Material¶\n\n    Nova CLI Command References: the complete command reference for all the daemons and admin tools that come with nova.\n    Configuration:\n        .. toctree::\n   :maxdepth: 1\n\n   configuration/index","accounts_in_message":[],"_revision_number":3},{"id":"3575ccb9da2c91cf817d12775048c1f836aee15c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-30 11:37:46.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/498818/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"17cbecb589dd37d23e30770806aa8e6d0e30ee15","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-08-30 11:43:23.000000000","message":"Patch Set 3:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":3},{"id":"561e28a562e62b4ca70cf62db5881f476b6b29b8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 11:47:09.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c86292b : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6db260f87572df54e58ccf0ce39fb97d551afdb9","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-08-30 11:51:15.000000000","message":"Patch Set 3:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":3},{"id":"672f0f3453d58379c41e1f184120c627ac4fe7af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 11:54:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/3/check/gate-nova-docs-ubuntu-xenial/116ba25//doc/build/html/ : SUCCESS in 5m 06s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/3/check/gate-nova-pep8-ubuntu-xenial/61e01bf/ : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9623e889bd163148e5f4f1d4fce647f4dc5817e2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-30 12:56:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dae50d1fca0963cf2b30fb49d3778f6435f6ac37","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-30 13:10:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498818/3/ : SUCCESS in 2h 21m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7cf5880a04bab8b71715501b89799f86884a6997","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 13:34:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/3/check/tempest-dsvm-full-nfv-xenial/ffe9a73 : SUCCESS in 2h 19m 17s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/3/check/tempest-dsvm-intel-nfv-xenial/6189076 : SUCCESS in 50m 14s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/45fc3a3 : SUCCESS in 1h 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e430e88abec1b619232e8bec6f09d1b5e4fb4d64","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-30 15:51:27.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/498818/3/check/nova-out-of-tree-pvm/347147a : FAILURE in 5h 00m 46s\n- nova-in-tree-pvm http://184.172.12.213/18/498818/3/check/nova-in-tree-pvm/e5bb3bb : FAILURE in 5h 00m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7a91ae0230015a0038ba205091a7ca864fcce413","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-31 04:26:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/1/check/tempest-dsvm-full-nfv-xenial/3453e78 : SUCCESS in 2h 24m 58s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/1/check/tempest-dsvm-intel-nfv-xenial/3cc5319 : SUCCESS in 50m 53s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498818/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9b8aa73 : SUCCESS in 1h 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ad131bade9c222618cf23eb061da1ef7b04b68b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-09-01 01:12:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8608c8807fecd2e88ea7117a57397d570f9ad873","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-04 10:22:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f7f10d2147ea5347d4b98e9eee31dcb8f6b7fa9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 10:35:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/4/check/gate-nova-docs-ubuntu-xenial/7876e9c//doc/build/html/ : SUCCESS in 4m 35s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/4/check/gate-nova-pep8-ubuntu-xenial/2dbeef3/ : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":4},{"id":"1ad3be1578be0369210fcfc7278e04fc13895b0a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-04 10:42:29.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/142e08e : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":4},{"id":"af2da76f55383e990c32bf9de88e64e8a407ef22","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-04 11:34:44.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/498818/4/check/nova-out-of-tree-pvm/b60670f : FAILURE in 1h 10m 44s\n- nova-in-tree-pvm http://184.172.12.213/18/498818/4/check/nova-in-tree-pvm/a8ce711 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c55a6b6e2a2e62c2925ed8c5be7416f69e98c59e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-04 12:01:08.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-498818-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"c66bdbfb64e5dd1d581da3df6125e2e2067ed839","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-04 13:42:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498818/4/ : SUCCESS in 2h 18m 07s","accounts_in_message":[],"_revision_number":4},{"id":"962ed095abfc5733500cf6de8bee591f13fd2673","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-04 20:36:10.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/498818/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7b8f5bc795683d1addab7658229bd2d7cd33c5eb","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-09-05 06:40:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6884bd7fa0943037244a573ca89623533c3f0d47","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2017-09-06 11:50:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cae6d8857cf7f4316cb2aeac86ea7187944e4db4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-06 13:41:24.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"86e94155f9a9fcd6eff58d3aa94052bb7cb9faeb","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-06 13:47:21.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498818/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7734ab4 : FAILURE in 1m 56s","accounts_in_message":[],"_revision_number":5},{"id":"5a22ee574e6efc61057551a86124ad370f7550bd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 15:13:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/5/check/gate-nova-docs-ubuntu-xenial/6921ba9//doc/build/html/ : SUCCESS in 4m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/5/check/gate-nova-pep8-ubuntu-xenial/907c240/ : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":5},{"id":"2ce0e179ed9f59b8adcd371c77b3f607c43144f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-06 16:50:26.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-498818-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"302a3429693c84b797ea634d32f8c6d05f166c4f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-06 17:33:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498818/5/ : SUCCESS in 2h 17m 44s","accounts_in_message":[],"_revision_number":5},{"id":"96d7fe5f094fa86e5fefc5137add1fd4bde7fc5e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-09-06 18:17:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nYup, easy win. ++\n\n-jay","accounts_in_message":[],"_revision_number":5},{"id":"19de67322b047abfda85c4adc62327b5479b0f20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 18:17:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"379c6d05602d1ab9b897181c75d534603174e644","date":"2017-09-06 19:08:30.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":5},{"id":"2c828b157bdaef378b92c1022360934c1e11a21d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 19:08:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/18/498818/5/gate/gate-nova-docs-ubuntu-xenial/00e8241//doc/build/html/ : SUCCESS in 5m 28s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/18/498818/5/gate/gate-nova-pep8-ubuntu-xenial/7091870/ : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":5},{"id":"545384f279957a9678b07646a41961e08407ec3f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-06 20:53:29.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/498818/5/check/nova-out-of-tree-pvm/2fa74e2 : SUCCESS in 2h 04m 23s\n- nova-in-tree-pvm http://184.172.12.213/18/498818/5/check/nova-in-tree-pvm/ced12d4 : FAILURE in 1h 06m 42s","accounts_in_message":[],"_revision_number":5},{"id":"a3edd2a5fe6d5cc6f17280982f82fdb62ffb61a8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-07 18:03:32.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/498818/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"e95592af68d49bc1ea90f6a038892b0c45be7b64","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-04 00:38:43.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 694d264bb82e6cf2573c601fbca06de0e57393c0","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2dedda8b6b583c6eed9dddd4963297eea04a064b","revisions":{"6dbbe4a53689c86cf20c92d9fde24078b2e6e865":{"kind":"REWORK","_number":1,"created":"2017-08-29 14:07:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/498818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/498818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/498818/1"}}},"commit":{"parents":[{"commit":"123a98caafc770d2b201c9d04348785221dea539","subject":"doc: Add user index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/123a98caafc770d2b201c9d04348785221dea539"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-24 14:58:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 13:52:35.000000000","tz":60},"subject":"doc: Add configuration index page","message":"doc: Add configuration index page\n\nChange-Id: Ic318e04be73639de7d26e53e1f0b152d9803fd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6dbbe4a53689c86cf20c92d9fde24078b2e6e865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6dbbe4a53689c86cf20c92d9fde24078b2e6e865"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"123a98caafc770d2b201c9d04348785221dea539","is_merged_in_target_branch":false,"change_id":"I160fed0f1b507eba7958666bf38f4138b0f62b3a","change_number":498817,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3d0f01e4bc032daa4a9d0911a2341d6664398780":{"kind":"REWORK","_number":2,"created":"2017-08-29 16:18:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/498818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/498818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/498818/2"}}},"commit":{"parents":[{"commit":"536b23746900bfd2d2359ff03d32ddb60706af94","subject":"doc: Add user index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/536b23746900bfd2d2359ff03d32ddb60706af94"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-24 14:58:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:09:12.000000000","tz":60},"subject":"doc: Add configuration index page","message":"doc: Add configuration index page\n\nChange-Id: Ic318e04be73639de7d26e53e1f0b152d9803fd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d0f01e4bc032daa4a9d0911a2341d6664398780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d0f01e4bc032daa4a9d0911a2341d6664398780"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"536b23746900bfd2d2359ff03d32ddb60706af94","is_merged_in_target_branch":false,"change_id":"I160fed0f1b507eba7958666bf38f4138b0f62b3a","change_number":498817,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"eb3e3a3cb1eed33af3e948160007ced5c09454fb":{"kind":"REWORK","_number":3,"created":"2017-08-30 10:47:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/498818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/498818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/498818/3"}}},"commit":{"parents":[{"commit":"e275e7f7f59721a32192558924468782e959545a","subject":"doc: Add user index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e275e7f7f59721a32192558924468782e959545a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-24 14:58:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-30 08:37:44.000000000","tz":60},"subject":"doc: Add configuration index page","message":"doc: Add configuration index page\n\nChange-Id: Ic318e04be73639de7d26e53e1f0b152d9803fd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb3e3a3cb1eed33af3e948160007ced5c09454fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb3e3a3cb1eed33af3e948160007ced5c09454fb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e275e7f7f59721a32192558924468782e959545a","is_merged_in_target_branch":false,"change_id":"I160fed0f1b507eba7958666bf38f4138b0f62b3a","change_number":498817,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e306d59c9856ca08cdb47532597a5d66133b9a3b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-09-04 10:22:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/498818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/498818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/498818/4"}}},"commit":{"parents":[{"commit":"13f73b6a8ab32003ae29ef1c8c02547e9f98c2ee","subject":"doc: Add user index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13f73b6a8ab32003ae29ef1c8c02547e9f98c2ee"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-24 14:58:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-04 10:20:38.000000000","tz":60},"subject":"doc: Add configuration index page","message":"doc: Add configuration index page\n\nChange-Id: Ic318e04be73639de7d26e53e1f0b152d9803fd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e306d59c9856ca08cdb47532597a5d66133b9a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e306d59c9856ca08cdb47532597a5d66133b9a3b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"13f73b6a8ab32003ae29ef1c8c02547e9f98c2ee","is_merged_in_target_branch":false,"change_id":"I160fed0f1b507eba7958666bf38f4138b0f62b3a","change_number":498817,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2dedda8b6b583c6eed9dddd4963297eea04a064b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2017-09-06 13:41:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/498818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/498818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/498818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/498818/5"}}},"commit":{"parents":[{"commit":"1eb53dfbcd6480691d8d3b80a6096c8be66560c5","subject":"doc: Add user index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1eb53dfbcd6480691d8d3b80a6096c8be66560c5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-24 14:58:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-06 13:40:17.000000000","tz":60},"subject":"doc: Add configuration index page","message":"doc: Add configuration index page\n\nChange-Id: Ic318e04be73639de7d26e53e1f0b152d9803fd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dedda8b6b583c6eed9dddd4963297eea04a064b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dedda8b6b583c6eed9dddd4963297eea04a064b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1eb53dfbcd6480691d8d3b80a6096c8be66560c5","is_merged_in_target_branch":true,"change_id":"I160fed0f1b507eba7958666bf38f4138b0f62b3a","change_number":498817,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
