)]}'
{"id":"openstack%2Fnova~498819","triplet_id":"openstack%2Fnova~master~I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3","project":"openstack/nova","branch":"master","topic":"bp/doc-migration","hashtags":[],"change_id":"I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3","subject":"doc: Cleanup of existing index pages","status":"ABANDONED","created":"2017-08-29 14:07:14.000000000","updated":"2020-03-18 10:16:16.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d02fe1806d5689e73b5a75a5f573e17ca82366c","_number":498819,"virtual_id_number":498819,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-17 20:16:12.000000000","_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"date":"2017-10-09 10:27:39.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-10 11:30:36.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-09 10:09:15.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-10-10 11:31:00.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-17 20:16:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2017-11-25 05:49:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-03-17 20:16:37.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-17 20:16:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"date":"2017-10-09 10:53:59.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2017-10-10 10:43:07.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"date":"2017-10-09 13:02:31.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":-1,"date":"2017-10-19 13:53:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-05 06:45:44.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-09-12 05:24:05.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 10:09:15.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-09 10:27:39.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 10:53:59.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-09 13:02:31.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 10:43:07.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-10-10 11:30:36.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-10 11:31:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 13:53:47.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-11-25 05:49:49.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-17 20:16:12.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2020-03-17 20:16:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-17 20:16:37.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-17 20:16:59.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21f4bc6d768f98b0cf6ff5792bb9f92279d0a3cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-29 14:07:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec10a8e87cd549bad0cc0487f470e90adaa86faa","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 14:08:26.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9f50617 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"d42b64c11ffee54757d611f98f3dde71e604a03b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 14:22:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/1/check/gate-nova-docs-ubuntu-xenial/e3b93c7//doc/build/html/ : SUCCESS in 7m 09s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/1/check/gate-nova-pep8-ubuntu-xenial/de1c0cd/ : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"452da1986071eb85cfa3a7c89b085be9e3a98566","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-29 14:23:31.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"2bfd1c3cbbeaf45be63fcd32a855ae56b35af215","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-29 15:44:24.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/1/check/nova-out-of-tree-pvm/5823262 : SUCCESS in 1h 35m 59s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/1/check/nova-in-tree-pvm/d5f306c : FAILURE in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3dfc562a0dc80202b010778af23ba02e84bfdba1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 16:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/1/ : SUCCESS in 2h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e2cd7c1622be734c73186eb8627b7a21ca11eebe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-29 16:18:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"079167940fa1ccf37ca3210ff325e8c81dbebe2e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 16:34:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://logs.openstack.org/19/498819/2/check/gate-nova-docs-ubuntu-xenial/5702712/ : FAILURE in 9m 21s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/2/check/gate-nova-pep8-ubuntu-xenial/2074b90/ : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f60912865ae7df8d3c3c572f00afe75af7638ed7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-29 17:49:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/2/check/nova-out-of-tree-pvm/0015c5b : SUCCESS in 1h 29m 41s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/2/check/nova-in-tree-pvm/f3593f1 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":2},{"id":"836ca0a25c6273d41ee95a5ef62b1fb232005e79","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-29 18:25:03.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/498819/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"c1129f5393047785cb21c82066209e9bcab9ec64","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 18:46:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498819/2/check/tempest-dsvm-full-nfv-xenial/9178828 : SUCCESS in 2h 24m 54s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498819/2/check/tempest-dsvm-intel-nfv-xenial/6fdf717 : SUCCESS in 49m 58s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-29/498819/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/4817f04 : SUCCESS in 1h 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a88aa18d19d74db3896ef67ee91d51e6ec9fa4d2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-29 19:15:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/2/ : SUCCESS in 2h 02m 46s","accounts_in_message":[],"_revision_number":2},{"id":"548fe478a2b512fdf32cd32ad1c5f4a6f6655105","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-29 19:50:38.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d764f92 : FAILURE in 2h 26m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6bb2236cf55e476fe2c6bb8adb4c0ab558f15ddb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-08-30 10:47:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"823cae2f35a4902c88539370165fc4e2c561adec","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 10:52:36.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f1bc621 : FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"20c7935f9bff256a84b81d053176d5298e32cd94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 11:00:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/3/check/gate-nova-docs-ubuntu-xenial/4033031//doc/build/html/ : SUCCESS in 4m 50s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/3/check/gate-nova-pep8-ubuntu-xenial/1471931/ : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":3},{"id":"0cd6ffbc79cb64222f0f88d5370fa4256024c1d4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-08-30 11:01:38.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"3a4ed223e15bf9155acc000a2485b0abddf7864e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-30 11:37:42.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/498819/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2e460dd2a88a2e37db259a29d7242492df0569a2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-30 13:02:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/3/ : SUCCESS in 2h 13m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b3336fa6216d905093251f6aa9db60f310158af9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-08-30 13:19:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/3/check/nova-out-of-tree-pvm/47bf1c5 : SUCCESS in 2h 30m 05s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/3/check/nova-in-tree-pvm/b79c3f1 : SUCCESS in 1h 29m 39s","accounts_in_message":[],"_revision_number":3},{"id":"09d2581cd706e28724170ae729cd80f50de8477c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-30 13:42:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498819/3/check/tempest-dsvm-full-nfv-xenial/a04e106 : SUCCESS in 2h 26m 48s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498819/3/check/tempest-dsvm-intel-nfv-xenial/5450c31 : SUCCESS in 48m 36s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-30/498819/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/247aaeb : SUCCESS in 1h 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef5e36dc41271d23521471efb9a8786b69f4319d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-08-31 04:26:58.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"bd05435a1ff707b5c8404b06b3fbbc7998aabb4e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 07:40:37.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"3f672641fe79c0a572511cc699763c40ecc646a8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-04 10:22:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f7a74b26c1ed26c153badd3aa373b0d41464f25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-04 10:35:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/4/check/gate-nova-docs-ubuntu-xenial/897ffa4//doc/build/html/ : SUCCESS in 4m 31s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/4/check/gate-nova-pep8-ubuntu-xenial/5b7acfe/ : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6e77bea71e8df856fc8eca62cb016015a154903a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-04 10:42:08.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2b6a426 : FAILURE in 16m 13s","accounts_in_message":[],"_revision_number":4},{"id":"5106207284db7fd0a4d465008d0964eafc9d4fca","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-04 11:59:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/4/check/nova-out-of-tree-pvm/08211d4 : SUCCESS in 1h 35m 43s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/4/check/nova-in-tree-pvm/1786ab2 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3f519a912e78d2225915f7744987be2472bb1f48","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-04 13:00:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/4/ : SUCCESS in 2h 10m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0c17b0bf0b2834fac79b46a97504bab543f73f42","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-04 13:21:23.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"1cf88229d809980e9a3e07c735ebda8fcc603c99","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-04 19:50:38.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/498819/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9367c98d3749488fb3753cec93cab68612c47ec1","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-09-05 06:45:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"803a6394b7a83b167f0fb64655245ff551afa2aa","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2017-09-06 11:51:50.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c9a66f05b821c377186ff5cc06cebe842482be4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-06 13:41:24.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4c62fd46d16992121d7ce292d433f7a7f367494b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-06 13:47:02.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/be92619 : FAILURE in 1m 53s","accounts_in_message":[],"_revision_number":5},{"id":"17676784d1bb7b49113e8fa2c505f8e64e2a859d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 15:11:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/5/check/gate-nova-docs-ubuntu-xenial/587e2eb//doc/build/html/ : SUCCESS in 5m 11s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/5/check/gate-nova-pep8-ubuntu-xenial/69e8692/ : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":5},{"id":"974ac5755095dc65d3de12c8ffe239578e528457","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-06 16:30:53.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"c2f5a774e3058da7cc4b8db1f26dd18c9a46e7b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-06 17:01:59.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"acd0d62e15bb38effe25113bbe363b03446fd70f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-06 17:16:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/5/ : SUCCESS in 2h 13m 08s","accounts_in_message":[],"_revision_number":5},{"id":"4d651ca04313ca2da1c8f83d227b6158dbd75a6e","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-07 16:57:06.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/498819/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"4101c716518f2942d9549df762902c25aa12f268","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-08 15:50:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"daeed760cef96e4b3b503fa20bd9e31020debc9c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-08 17:58:36.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/6/check/nova-out-of-tree-pvm/1fa7834 : SUCCESS in 2h 07m 11s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/6/check/nova-in-tree-pvm/4592c07 : FAILURE in 36m 16s","accounts_in_message":[],"_revision_number":6},{"id":"a546dd95fce2d017c4783a82c011049877016e9a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-08 18:18:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/6/ : SUCCESS in 2h 26m 08s","accounts_in_message":[],"_revision_number":6},{"id":"d0ec3a4c0ef5bacce4e54a9188a7b933d4a4d92f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-08 19:57:33.000000000","message":"Patch Set 6:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9341109 : FAILURE in 4h 00m 33s","accounts_in_message":[],"_revision_number":6},{"id":"60cf49c26193730f904d54a8eaf49920e412e19a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-08 22:59:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/6/check/gate-nova-docs-ubuntu-xenial/45e05e7//doc/build/html/ : SUCCESS in 5m 06s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/6/check/gate-nova-pep8-ubuntu-xenial/cd4165f/ : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":6},{"id":"c4e7f8c85649a493bbacbfca5e964953cbdd56c0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-12 05:24:05.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/498819/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"d030f22dd43ad16ced3592b5ae8b60db4c803450","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-19 08:44:29.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"c15916ee20e1f49f4ee2350e245461a917882436","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-22 14:58:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"88339ea83aba2fd49fb5e14537d77a093b30e80f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 15:11:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/7/check/gate-nova-docs-ubuntu-xenial/4c7833d//doc/build/html/ : SUCCESS in 4m 44s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/7/check/gate-nova-pep8-ubuntu-xenial/9e191c1/ : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":7},{"id":"666a8d1f24991a1f0af079fe5d352f84166fb68c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-09-22 15:11:41.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"9fb88f47cc7624af6072487bf3b72797491fd4a8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-22 17:31:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/498819/7/ : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":7},{"id":"f88cd013e4cc6ec606cbd24a459dcf5ec486a230","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-22 17:37:50.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"32d07070ba9b9a1127b90b6edf6ed71b4e75621f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-22 17:50:32.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/7/check/nova-out-of-tree-pvm/7dad9f3 : FAILURE in 2h 51m 17s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/7/check/nova-in-tree-pvm/121dd34 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":7},{"id":"1331ae88d105cb93b41301f53a35d3d6a48201d0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-22 17:52:13.000000000","message":"Patch Set 7:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/8e579b6 : SUCCESS in 2h 00m 25s","accounts_in_message":[],"_revision_number":7},{"id":"d635730c16fccdd806c35c4c6801f7296591ddd9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-26 07:14:32.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"5e1e6b8a1c3a15f7f6ad6d482fa7c8a3e7f7cb58","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-09 09:55:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"836a7aab7e244f8dd0737b0d9ee8e0762a729312","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-09 10:09:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/8/check/gate-nova-docs-ubuntu-xenial/6fcf0c8//doc/build/html/ : SUCCESS in 11m 07s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/8/check/gate-nova-pep8-ubuntu-xenial/845327d/ : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":8},{"id":"6d62eaa8d4ccfd47ce2953038436f647fd63d31f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 10:09:39.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"eff6ac079a6ae5b3258ca37abb2e7fa66aaabbc5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 10:14:31.000000000","message":"Patch Set 8: Verified-1\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/53c8594 : FAILURE in 18m 07s","accounts_in_message":[],"_revision_number":8},{"id":"95be06942aab98c53ad74ebd4bf3d8a2a01230b2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-09 10:27:39.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/19/498819/8/check/nova-out-of-tree-pvm/9fba8b2 : FAILURE in 31m 02s\n- nova-in-tree-pvm http://184.172.12.213/19/498819/8/check/nova-in-tree-pvm/7df664a : FAILURE in 29m 42s","accounts_in_message":[],"_revision_number":8},{"id":"77b0ce722699ae89056dc11872c0c8a5307214dc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-09 10:53:59.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-19-498819-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"709dbf57998474c7a0ddb750dc3db7436bb43c17","author":{"_account_id":11351,"name":"Andreas Maier","email":"maiera@de.ibm.com","username":"maiera"},"date":"2017-10-09 12:07:05.000000000","message":"Patch Set 8:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":8},{"id":"50659183c0d2e787aa1be1dc712b1825032ee805","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-09 12:30:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/8/check/gate-nova-docs-ubuntu-xenial/af22047//doc/build/html/ : SUCCESS in 5m 08s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/8/check/gate-nova-pep8-ubuntu-xenial/9b4a07b/ : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":8},{"id":"9a6976551fd2587cd8939ea66c2351c01d96683b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 12:31:14.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"44062005167e08680c441614d7022db1b8a014b3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 13:02:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline). To recheck leave a comment with \u0027intel-nfv-ci recheck\u0027. For more info go to https://wiki.openstack.org/wiki/ThirdPartySystems/Intel_NFV_CI.\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-09/498819/8/check/tempest-dsvm-full-nfv-xenial/9464fa5 : FAILURE in 1h 52m 22s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-09/498819/8/check/tempest-dsvm-intel-nfv-xenial/ee61cbd : FAILURE in 36m 13s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-10-09/498819/8/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9070cd4 : FAILURE in 3h 06m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ab0c095b0bce7ce353a02635e513260ea20de283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-09 13:16:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/19/498819/8/check/openstack-tox-pep8/b49433c/ : POST_FAILURE in 8m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/19/498819/8/check/build-openstack-sphinx-docs/9e209bc/html/ : SUCCESS in 4m 46s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/19/498819/8/check/legacy-tempest-dsvm-py35/f4e8bb9/ : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":8},{"id":"bc751966f36dfe0c49ddee8ac7b28b1d8f9c68b7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-09 13:52:16.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/498819/8/ : FAILURE in 3h 55m 57s","accounts_in_message":[],"_revision_number":8},{"id":"52161adce0521b7db3fe27e6cb0ecd816337e6fc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 15:57:27.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/880112e : FAILURE in 2h 00m 07s","accounts_in_message":[],"_revision_number":8},{"id":"7013126f4224d8c22718c1ae2b1c66edbbda2688","author":{"_account_id":11351,"name":"Andreas Maier","email":"maiera@de.ibm.com","username":"maiera"},"date":"2017-10-09 17:30:01.000000000","message":"Patch Set 8:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":8},{"id":"e49f2850cdd13ba69fc1ff737119d098c1e04713","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-09 20:45:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/8/check/gate-nova-docs-ubuntu-xenial/4647841//doc/build/html/ : SUCCESS in 5m 08s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/8/check/gate-nova-pep8-ubuntu-xenial/a1ff1fb/ : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":8},{"id":"61b1b2fe633f1dc395cdb456a7a73791731a685f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 20:46:12.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"a9ac579237d0eaed27df27d9f4d69f0cef399ceb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-09 22:45:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/19/498819/8/check/openstack-tox-pep8/6bda143/ : SUCCESS in 9m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/19/498819/8/check/build-openstack-sphinx-docs/63c7823/html/ : SUCCESS in 5m 09s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/19/498819/8/check/legacy-tempest-dsvm-py35/a23cbc2/ : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":8},{"id":"64fa725e59f3c5c30a526d4cf768a8ef44aa0610","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-09 22:47:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"8294a9a9b4999b4209307cad2017e45a045fa3df","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 03:06:27.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/10499bc : FAILURE in 2h 26m 18s","accounts_in_message":[],"_revision_number":8},{"id":"8293f1cce9ffea6fbe5980b9cc03eb7b3939f8bb","author":{"_account_id":11351,"name":"Andreas Maier","email":"maiera@de.ibm.com","username":"maiera"},"date":"2017-10-10 06:30:30.000000000","message":"Patch Set 8:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":8},{"id":"96d0b2e60b4cbbc53032a12162e03ffdd1ac06c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-10 06:49:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/19/498819/8/check/gate-nova-docs-ubuntu-xenial/b894be6//doc/build/html/ : SUCCESS in 8m 30s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/19/498819/8/check/gate-nova-pep8-ubuntu-xenial/36c68fb/ : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":8},{"id":"1520a6629349bee95ec1ec1992df0a0a3eeb39e4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 06:49:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"dbcfc0ae6bba64016906586c650c661f0a783cf6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-10 10:43:07.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/498819/8/ : FAILURE in 3h 38m 37s","accounts_in_message":[],"_revision_number":8},{"id":"d70862c6632fe20a6414894a398d72e7ad941ee3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 11:30:36.000000000","message":"Patch Set 8: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/498819/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f5566ff : SUCCESS in 1h 56m 47s","accounts_in_message":[],"_revision_number":8},{"id":"50631483ae263ab9919ecfe2d230293d4358659f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-10 11:31:00.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":8},{"id":"de6d9057304c6f20e0f118b9f1a1ee252754e0f6","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-10-19 13:53:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3c23dd74fe8d98c25255602c75020b7907ef21e6","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2017-11-25 04:37:11.000000000","message":"Patch Set 8:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":8},{"id":"fb610edc2a3012a751b77425ba1842289d7bc310","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-25 05:49:49.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-9578 : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":8},{"id":"4edb1eab1b2b55f139e50babcc0662ec9be4a4c1","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-17 20:15:49.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"96cce0b46bcead4fba66932cb0d7224409774729","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-17 20:16:12.000000000","message":"Patch Set 8:\n\nGrr, wanted to see the docs preview site, didn\u0027t notice this was in merge conflict :(","accounts_in_message":[],"_revision_number":8},{"id":"d4c73265eaddbd2ab2be83cfad79eaea5c85da9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:16:16.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"3efdb8672f995d1dcc139cc712dbeefec43ef25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:16:21.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"159807dd4f4dc718e7158b875168b45f99de9ca7","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-17 20:16:37.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"093ffc3176949ca0bfe60401695bcd9486043beb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-17 20:16:59.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"05a6ce729bb7165e90d3597ccb4d3b9fed6ef7d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-18 10:16:16.000000000","message":"Abandoned\n\nThis needs to be done but I don\u0027t know when I\u0027ll be able to do it. Best abandon and recreate later","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8f3f819681c80eab53850cdc9b7e7ea1491ed0a0","revisions":{"451d34cd5131ff54a77cd0c318c43e79e95a4954":{"kind":"REWORK","_number":1,"created":"2017-08-29 14:07:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/1"}}},"commit":{"parents":[{"commit":"6dbbe4a53689c86cf20c92d9fde24078b2e6e865","subject":"doc: Add configuration index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6dbbe4a53689c86cf20c92d9fde24078b2e6e865"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 13:58:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 13:58:24.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/451d34cd5131ff54a77cd0c318c43e79e95a4954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/451d34cd5131ff54a77cd0c318c43e79e95a4954"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6dbbe4a53689c86cf20c92d9fde24078b2e6e865","is_merged_in_target_branch":false,"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","change_number":498818,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ca2ef2e893f7d2813ff33060a87cdb4cf25088fa":{"kind":"REWORK","_number":2,"created":"2017-08-29 16:18:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/2"}}},"commit":{"parents":[{"commit":"3d0f01e4bc032daa4a9d0911a2341d6664398780","subject":"doc: Add configuration index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d0f01e4bc032daa4a9d0911a2341d6664398780"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:17:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:17:07.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca2ef2e893f7d2813ff33060a87cdb4cf25088fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca2ef2e893f7d2813ff33060a87cdb4cf25088fa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3d0f01e4bc032daa4a9d0911a2341d6664398780","is_merged_in_target_branch":false,"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","change_number":498818,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4129a1f27176aac99f71142b6f947e1be36b1f33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-08-30 10:47:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/3"}}},"commit":{"parents":[{"commit":"eb3e3a3cb1eed33af3e948160007ced5c09454fb","subject":"doc: Add configuration index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb3e3a3cb1eed33af3e948160007ced5c09454fb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:17:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-30 08:37:47.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4129a1f27176aac99f71142b6f947e1be36b1f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4129a1f27176aac99f71142b6f947e1be36b1f33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb3e3a3cb1eed33af3e948160007ced5c09454fb","is_merged_in_target_branch":false,"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","change_number":498818,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8796434235c516af0a8e2f6ac31351dca673dfea":{"kind":"REWORK","_number":4,"created":"2017-09-04 10:22:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/4"}}},"commit":{"parents":[{"commit":"e306d59c9856ca08cdb47532597a5d66133b9a3b","subject":"doc: Add configuration index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e306d59c9856ca08cdb47532597a5d66133b9a3b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:17:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-04 10:22:14.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8796434235c516af0a8e2f6ac31351dca673dfea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8796434235c516af0a8e2f6ac31351dca673dfea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e306d59c9856ca08cdb47532597a5d66133b9a3b","is_merged_in_target_branch":false,"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","change_number":498818,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ba3fa68982a5555f6e7b8b66e213855ffd6e45b8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2017-09-06 13:41:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/5"}}},"commit":{"parents":[{"commit":"2dedda8b6b583c6eed9dddd4963297eea04a064b","subject":"doc: Add configuration index page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dedda8b6b583c6eed9dddd4963297eea04a064b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-08-29 16:17:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-06 13:40:17.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba3fa68982a5555f6e7b8b66e213855ffd6e45b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba3fa68982a5555f6e7b8b66e213855ffd6e45b8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2dedda8b6b583c6eed9dddd4963297eea04a064b","is_merged_in_target_branch":true,"change_id":"Ic318e04be73639de7d26e53e1f0b152d9803fd2f","change_number":498818,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b39054bd83040d1f2926a2344201ee6d6a6bc74e":{"kind":"REWORK","_number":6,"created":"2017-09-08 15:50:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/6"}}},"commit":{"parents":[{"commit":"28cf7b167ff3743d0dbfcd298973befa8e23e3ab","subject":"Merge \"Typo error about help resource_classes.inc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28cf7b167ff3743d0dbfcd298973befa8e23e3ab"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 15:50:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 15:50:18.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b39054bd83040d1f2926a2344201ee6d6a6bc74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b39054bd83040d1f2926a2344201ee6d6a6bc74e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"28cf7b167ff3743d0dbfcd298973befa8e23e3ab","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"20d387c8c9a3937c7eb0cbadd017787998a7be19":{"kind":"REWORK","_number":7,"created":"2017-09-22 14:58:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/7"}}},"commit":{"parents":[{"commit":"2f36a355f29cb9f23beb2b80399e59f02d3c17a3","subject":"docs: Rename cellsv2_layout -\u003e cellsv2-layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f36a355f29cb9f23beb2b80399e59f02d3c17a3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 15:50:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-22 14:56:52.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20d387c8c9a3937c7eb0cbadd017787998a7be19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20d387c8c9a3937c7eb0cbadd017787998a7be19"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2f36a355f29cb9f23beb2b80399e59f02d3c17a3","is_merged_in_target_branch":true,"change_id":"I5c99ff6b04ee97bac210a0d6762015225775c5ee","change_number":498821,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8f3f819681c80eab53850cdc9b7e7ea1491ed0a0":{"kind":"REWORK","_number":8,"created":"2017-10-09 09:55:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/498819/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/19/498819/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/19/498819/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/19/498819/8"}}},"commit":{"parents":[{"commit":"79ce4bef483b72a4e4e67c3e1b09d7a4a3922094","subject":"Merge \"Set migration object attributes for source/dest during live migrate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79ce4bef483b72a4e4e67c3e1b09d7a4a3922094"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 15:50:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-10-09 09:53:43.000000000","tz":60},"subject":"doc: Cleanup of existing index pages","message":"doc: Cleanup of existing index pages\n\nCross-reference documents where appropriate and fix some formatting.\n\nChange-Id: I6b6448cec3f9d0bd467e4405a6d72d069efa0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f3f819681c80eab53850cdc9b7e7ea1491ed0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f3f819681c80eab53850cdc9b7e7ea1491ed0a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"79ce4bef483b72a4e4e67c3e1b09d7a4a3922094","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
