)]}'
{"id":"openstack%2Fnova~499826","triplet_id":"openstack%2Fnova~master~I6a1d320ce914926791d5f45e89bf4c601a6b10a0","project":"openstack/nova","branch":"master","topic":"bug/1714275","hashtags":[],"change_id":"I6a1d320ce914926791d5f45e89bf4c601a6b10a0","subject":"Include /resource_providers/uuid/allocations link","status":"MERGED","created":"2017-08-31 22:07:28.000000000","updated":"2017-10-27 17:45:29.000000000","submitted":"2017-10-27 17:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"499826-1509126329541-9ea1339d","meta_rev_id":"08c2284cc8d4daa7a511fbbec6584beafa50b331","_number":499826,"virtual_id_number":499826,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-10-23 20:15:08.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-23 20:10:23.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-23 20:18:59.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-23 22:10:28.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-23 20:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-23 19:58:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2017-10-27 17:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-23 21:09:42.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-23 21:45:08.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"date":"2017-10-23 23:18:16.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2017-10-25 16:35:56.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2017-10-25 18:46:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":2,"date":"2017-10-27 15:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2017-10-24 02:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":1,"date":"2017-10-27 15:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-01 14:22:42.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2017-09-01 15:36:24.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-01 16:34:17.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-01 16:49:41.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-01 16:59:41.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-01 17:50:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-04 05:13:23.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2017-10-12 21:26:35.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2017-10-23 09:46:51.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2017-10-23 13:08:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 17:27:18.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2017-10-23 19:58:03.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-10-23 20:04:08.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 20:10:23.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 20:15:08.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 20:18:59.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 21:09:42.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 21:45:08.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-10-23 22:10:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-23 23:18:16.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-24 02:02:40.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-10-25 16:35:56.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-10-25 18:46:22.000000000","updated_by":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"reviewer":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"state":"REVIEWER"},{"updated":"2017-10-27 15:07:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2017-10-27 17:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45a250ba68b9da7381995e44476b3ceec46ee9dd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-08-31 22:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12b6e7e82203ceddfc5a8530837a544e84d3da2a","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-08-31 22:13:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5512770b23babcd00c8efbaeabc5d7507b5fb2c2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-08-31 22:33:04.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/499826/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b842aae23d1554b2c9d2be2c6f938d19eec7a31b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-08-31 23:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/1/check/tempest-dsvm-full-xenial/4f7b3be/ : SUCCESS in 1h 14m 10s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/1/check/grenade-dsvm-xenial/39c8579/ : SUCCESS in 50m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2e510336c766ae18329802f2b1fa08fba46e2bb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-08-31 23:31:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/1/ : FAILURE in 1h 23m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1cc89fd57ce38e079b7a1c0adc2bbb8e6f160ae3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-31 23:56:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/499826/1/check/dsvm-tempest-neutron-network/e63441b : SUCCESS in 1h 48m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1490062e4b88dff7f3085ee8579b0068a85d1f97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-31 23:58:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/26/499826/1/check/gate-nova-docs-ubuntu-xenial/c6594c9//doc/build/html/ : SUCCESS in 5m 01s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-nova-pep8-ubuntu-xenial/4514c53/ : SUCCESS in 9m 14s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-nova-python27-ubuntu-xenial/af3bfa9/ : SUCCESS in 12m 20s\n- gate-nova-python35 http://logs.openstack.org/26/499826/1/check/gate-nova-python35/18feb0b/ : SUCCESS in 16m 17s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/c46d238/ : SUCCESS in 1h 32m 23s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/cfc8f24/ : SUCCESS in 55m 55s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/f5b9459/ : SUCCESS in 1h 24m 45s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-nova-tox-functional-ubuntu-xenial/18abcaf/ : FAILURE in 30m 27s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-nova-tox-functional-py35-ubuntu-xenial/82770e5/ : SUCCESS in 21m 33s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/8ce4555/ : SUCCESS in 1h 04m 34s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/26/499826/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/3b98022/ : FAILURE in 1h 01m 56s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-cells-ubuntu-xenial/41e537a/ : SUCCESS in 1h 19m 59s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/ead2d34/ : SUCCESS in 1h 07m 27s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/5c2f7c9/ : SUCCESS in 1h 42m 11s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/29198d2/ : SUCCESS in 1h 19m 24s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/ba5bfd7/ : SUCCESS in 47m 54s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/26/499826/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/81167fa/ : SUCCESS in 46m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1eb9073034ff48a4ad5faba9c2eec0d86932b796","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-01 00:06:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/26/499826/1/check/nova-out-of-tree-pvm/5439b38 : SUCCESS in 1h 58m 59s\n- nova-in-tree-pvm http://184.172.12.213/26/499826/1/check/nova-in-tree-pvm/d3e479a : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1681225ea88e3471c94e51c26625203b01b8b63e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-09-01 00:17:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-31/499826/1/check/tempest-dsvm-full-nfv-xenial/8e58728 : SUCCESS in 2h 09m 30s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-31/499826/1/check/tempest-dsvm-intel-nfv-xenial/519fabb : SUCCESS in 47m 48s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-08-31/499826/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/ecce404 : SUCCESS in 1h 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25a969bb9cd3cab52812f90b679df91b593e0fb6","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-01 00:22:30.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/499826/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f5c30c2 : FAILURE in 2h 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"56933de193b52148c96520e4d7c28c6353184a06","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-01 01:02:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/499826/1/check/dsvm-tempest-xen/c33e9a7 : SUCCESS in 2h 54m 10s","accounts_in_message":[],"_revision_number":1},{"id":"39ad16eb3f2faab0524b960ed9f54e0b6a7b9ed2","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-01 03:16:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/26/499826/1/check-nova/Nova-ML2-Sriov/dbdd127 : SUCCESS in 1h 06m 52s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/26/499826/1/check-nova/Nova-MACVTAP-ML2-Sriov/d2beb0d : SUCCESS in 43m 10s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"22238593dc9329aee3b63df89556cabe5d5737d0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-09-01 12:42:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2c7e2a1102b3a56217e4f476f45d81c354bd35b","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-09-01 12:56:58.000000000","message":"Patch Set 1:\n\nRegarding the need for a microversion, the guideline [0] states for X.Y versioning:\n\n\"Y should be changed when you make any change to the API. Note that this includes semantic changes which may not affect the input or output formats or even originate in the API code layer. We are not distinguishing between backwards compatible and backwards incompatible changes in the versioning system. It will however be made clear in the documentation as to what is a backwards compatible change and what is a backwards incompatible one.\"\n\n[0] https://specs.openstack.org/openstack/api-wg/guidelines/microversion_specification.html","accounts_in_message":[],"_revision_number":1},{"id":"8189cfb8f7b17907079edc86244275de9ffaec84","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-09-01 13:16:01.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThe strict interpretation is as Ed says, and we probably ought to go with that, but some comments within on how we might be able weasel around it if we want to.","accounts_in_message":[],"_revision_number":1},{"id":"f904b62ae24846b7f6643c01c8685355bcf26e47","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-09-01 13:22:29.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWho\u0027s the final arbiter here?","accounts_in_message":[],"_revision_number":1},{"id":"f83233b4df0be84fc91cb24faaf2020de3cfbac3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-09-01 13:23:49.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"5309117f8aee8717b25d57709d5311c8561a6ad6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-09-01 13:32:32.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027m happy to go ahead and make this change, because of the reasons both Eric and I have stated. I think it is the pragmatic thing to do. I\u0027m happy to be a purist about lots of stuff, but not so much with microversions, especially in cases of missing info.","accounts_in_message":[],"_revision_number":2},{"id":"9bdb828dc2c495b581c3c9bfe2aef78315f6f503","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-01 13:43:09.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/2/ : FAILURE in 15s","accounts_in_message":[],"_revision_number":2},{"id":"72c9bfe7e34eb1c63cb7d098e3caf582e33590a2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-09-01 14:05:32.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d2e887c1c089cab4ebc9f1b1a96f97ae1a8bc18a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-09-01 14:21:19.000000000","message":"Patch Set 3: Published edit on patch set 2","accounts_in_message":[],"_revision_number":3},{"id":"c4d0445c78c62e9f6cdf87acb7256259ff48a1cd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-01 14:21:27.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/499826/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"42d558bb57213ca827691957afa2d8edc2b6c41c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-09-01 14:21:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b58ba1f9ca333da5caccdd618462a9b7fcc86ffc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-09-01 15:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/3/check/tempest-dsvm-full-xenial/df44475/ : SUCCESS in 1h 10m 47s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/3/check/grenade-dsvm-xenial/097c3c6/ : SUCCESS in 51m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c25be6c592a666903e44469ac0ad5512c17907b0","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-01 15:36:24.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/499826/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"239ce612678fe14922d18f7d46f498c76b12bdb9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-01 15:39:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/26/499826/3/check/nova-out-of-tree-pvm/8144442 : SUCCESS in 1h 17m 35s\n- nova-in-tree-pvm http://184.172.12.213/26/499826/3/check/nova-in-tree-pvm/ee53123 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":3},{"id":"52ce53d996062aaf60ff2fb4b74232b8405a8e3e","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-09-01 16:34:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/26/499826/3/check/dsvm-tempest-xen/5b30012 : SUCCESS in 2h 07m 40s","accounts_in_message":[],"_revision_number":3},{"id":"6f67ad24a6e2555bffe6d7f439d393ee7c36c4cd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-09-01 16:44:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/499826/3/check/dsvm-tempest-neutron-network/4949e17 : SUCCESS in 2h 19m 47s","accounts_in_message":[],"_revision_number":3},{"id":"5429672b37f8292494aa5338399fdb808fc6b4ff","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-09-01 16:48:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2bd96edb3dc15ef6c540579f7cfe617eb7e8df44","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2017-09-01 16:49:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-full-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-09-01/499826/3/check/tempest-dsvm-full-nfv-xenial/d2d241d : SUCCESS in 2h 20m 56s\n- tempest-dsvm-intel-nfv-xenial http://intel-openstack-ci-logs.ovh/portland/2017-09-01/499826/3/check/tempest-dsvm-intel-nfv-xenial/f866b71 : SUCCESS in 49m 28s\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://intel-openstack-ci-logs.ovh/portland/2017-09-01/499826/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/ee3d176 : SUCCESS in 1h 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2569b9dc06c78e9c6ac0d6477069387ea0a1c42a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-01 16:59:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/26/499826/3/check/gate-nova-docs-ubuntu-xenial/5318d41//doc/build/html/ : SUCCESS in 4m 16s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-nova-pep8-ubuntu-xenial/6da2c81/ : SUCCESS in 9m 12s\n- gate-nova-python27-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-nova-python27-ubuntu-xenial/a04f291/ : SUCCESS in 12m 18s\n- gate-nova-python35 http://logs.openstack.org/26/499826/3/check/gate-nova-python35/82beb6e/ : SUCCESS in 14m 43s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/169112e/ : SUCCESS in 1h 32m 26s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3868bbe/ : SUCCESS in 52m 10s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/5e64928/ : SUCCESS in 1h 33m 14s\n- gate-nova-tox-functional-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-nova-tox-functional-ubuntu-xenial/5842562/ : SUCCESS in 24m 05s\n- gate-nova-tox-functional-py35-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-nova-tox-functional-py35-ubuntu-xenial/b425e77/ : SUCCESS in 19m 12s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/f06e3af/ : SUCCESS in 1h 02m 06s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/26/499826/3/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/63030fb/ : FAILURE in 55m 37s (non-voting)\n- gate-tempest-dsvm-cells-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-cells-ubuntu-xenial/7fc462b/ : SUCCESS in 1h 35m 21s\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/212f1e4/ : FAILURE in 1h 03m 39s (non-voting)\n- gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-neutron-multinode-full-ubuntu-xenial-nv/540f51d/ : SUCCESS in 1h 32m 49s (non-voting)\n- gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-neutron-nova-next-full-ubuntu-xenial-nv/2decd40/ : SUCCESS in 1h 25m 28s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/01a2e8d/ : SUCCESS in 40m 12s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/26/499826/3/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/b81bd33/ : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9c60e5b396954c8f242216111f18024a6455686","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-01 17:08:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/499826/3/ : SUCCESS in 2h 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d92f0610b69fa2eee3bc4d355999ba5ad85eb6df","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-09-01 17:10:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/26/499826/3/check-nova/Nova-ML2-Sriov/01d2f5e : FAILURE in 56m 38s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/26/499826/3/check-nova/Nova-MACVTAP-ML2-Sriov/edafbd2 : SUCCESS in 56m 28s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"3abd4be4cf996d1df3759d9c8b05cbad6eb05022","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-09-01 17:50:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/26/499826/3/check-vote/ext-nova-zuul/aeafc34 : FAILURE in 30m 29s","accounts_in_message":[],"_revision_number":3},{"id":"846148be9e22c1eda3bbc1c6293e7b563ec7d8bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-01 18:07:26.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-26-499826-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"eb3b02776f79b267cf9d5b9c4a1d5584439e0a69","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2017-09-04 05:13:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a19ebc2b13e439f257d44b591fff8064c8b8be40","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-09-09 02:23:38.000000000","message":"Patch Set 3: Code-Review+1\n\nI\u0027m still okay with this.","accounts_in_message":[],"_revision_number":3},{"id":"5c27fb99a60a7f3b438c6663df0c4dfb6bfcee40","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-10-12 21:26:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\n(4:24:21 PM) mriedem: the right thing to do is microversion the change, regardless of what wasn\u0027t done correctly before,\n(4:24:34 PM) mriedem: we bump the minimum required microversion that nova needs each release already,\n(4:24:42 PM) mriedem: so eventually the minimum we\u0027re going to require is already going to contain this thing,\n(4:24:51 PM) mriedem: so long-term it\u0027s not a big deal to do it in a new microversion,\n(4:24:53 PM) mriedem: so let\u0027s just do that\n(4:25:20 PM) mriedem: efried: ok?\n(4:25:49 PM) efried: Okay.","accounts_in_message":[],"_revision_number":3},{"id":"b606691ae528b5b62826c1cce72be76312275407","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-13 12:48:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a85ff74e660204b5b7a8e90a675fc7d53ea5e9b3","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-10-18 14:13:41.000000000","message":"Patch Set 3: Code-Review-1\n\nWhile I agree that there is no danger of the world exploding if this change is made without bumping the microversion, it really should. Using a prior failure as an example, IMO, only shows that we should be more vigilant, or else now we\u0027ll have two examples that future developers will cite when they don\u0027t want to bump the version because they feel that their change is \"compatible\".","accounts_in_message":[],"_revision_number":3},{"id":"2297195f49991fd3cc91886531195e24d670e059","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-20 20:15:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"74626b45277a8c05b8fcfaf3245b1c8205bb8f3a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-20 20:36:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e345fec7c6b468ff6512baaad911918454339f5b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-20 20:40:17.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f2bd3121f9cfcce53d2b788859bc3a1f2642ed88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-20 21:15:38.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/5/check/tempest-dsvm-full-xenial/c6ad234/ : FAILURE in 37m 57s","accounts_in_message":[],"_revision_number":5},{"id":"af8b3f3514f641d021d1607c104d2290fae780f9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-20 21:23:29.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/26/499826/5/check/nova-out-of-tree-pvm/e2a8a9e : FAILURE in 36m 49s\n- nova-in-tree-pvm http://184.172.12.213/26/499826/5/check/nova-in-tree-pvm/0a86c7b : FAILURE in 31m 43s","accounts_in_message":[],"_revision_number":5},{"id":"025979dc07aa26adc81bf7fef201446868357e24","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-20 21:24:18.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/5/check/check-dsvm-tempest-vz7-exe-minimal/f16fb11 : FAILURE in 45m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"90ac2d464463d7dfd05db43612faedae039d9b9d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-20 22:03:05.000000000","message":"Patch Set 5: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/499826/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/72bd679 : SUCCESS in 1h 25m 43s","accounts_in_message":[],"_revision_number":5},{"id":"d154b09d220f31d89881fbdfd9795fa4fe5cff06","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-20 22:03:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"806779bb0ed69dc24e97271a07b538be56b3010d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-20 22:14:29.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/499826/5/check/dsvm-tempest-neutron-network/48eb7c7 : FAILURE in 1h 36m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9ae85698866e2ee4de1345a769d2cc4fc5e3f069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-20 22:55:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/5/check/openstack-tox-pep8/043ecb6/ : SUCCESS in 8m 50s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/5/check/openstack-tox-py27/f8cad1c/ : POST_FAILURE in 11m 45s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/5/check/openstack-tox-py35/0c5cc02/ : POST_FAILURE in 5m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/5/check/build-openstack-sphinx-docs/b78fb81/html/ : SUCCESS in 5m 50s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-py35/8232d0b/ : FAILURE in 48m 35s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/5/check/build-openstack-releasenotes/7502da8/html/ : SUCCESS in 13m 12s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional/d92fcb8/ : FAILURE in 6m 23s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional-py35/70726a6/ : FAILURE in 7m 52s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode/43472be/ : SUCCESS in 1h 08m 30s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/7eef883/ : FAILURE in 1h 25m 33s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-cells/3f18bed/ : FAILURE in 37m 33s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/5951250/ : FAILURE in 50m 05s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-multinode-full/1844e93/ : FAILURE in 1h 04m 51s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-nova-next-full/6ae2b98/ : FAILURE in 52m 07s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-multinode-live-migration/3a644c9/ : FAILURE in 43m 46s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/38037ce/ : FAILURE in 42m 16s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-full/fcdd028/ : FAILURE in 55m 42s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron/ed7e635/ : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b6c0878a19d2409bdcc605f5abf4e506ef88fa9a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-20 23:20:11.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/5/ : FAILURE in 2h 42m 40s","accounts_in_message":[],"_revision_number":5},{"id":"a2c6e25969f3d42c85d5392d8f2a32eea839beaf","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-21 02:52:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/26/499826/5/check-nova/Nova-ML2-Sriov/15000f6 : FAILURE in 44m 49s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/26/499826/5/check-nova/Nova-MACVTAP-ML2-Sriov/992015e : FAILURE in 49m 27s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"fc0e5d58d7de93a3a52480e28ab279031c7bcc64","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-22 13:49:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6d58787543d03bfb1ca1f20549258bcfe2d086d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-22 14:29:37.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/5/check/tempest-dsvm-full-xenial/4515dde/ : FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d6589ee69fefe0d2c07f736a0383673754e3d471","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-22 14:45:36.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/5/check/check-dsvm-tempest-vz7-exe-minimal/edace18 : FAILURE in 55m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"6681cba91da88ccfed690992f32b4426b1a9ba8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-22 15:03:50.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/5/check/openstack-tox-pep8/170aa56/ : SUCCESS in 7m 16s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/5/check/openstack-tox-py27/3ec9c37/ : POST_FAILURE in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/5/check/openstack-tox-py35/e51ce7c/ : POST_FAILURE in 4m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/5/check/build-openstack-sphinx-docs/bda0dce/html/ : SUCCESS in 5m 10s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-py35/16c6886/ : FAILURE in 51m 24s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/5/check/build-openstack-releasenotes/faa5266/html/ : SUCCESS in 8m 10s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional/d7167f1/ : FAILURE in 6m 05s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional-py35/d5ee5ff/ : FAILURE in 9m 13s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode/556c401/ : SUCCESS in 1h 12m 26s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/ebf0cc1/ : FAILURE in 1h 00m 54s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-cells/6c3ba01/ : FAILURE in 36m 10s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3b1c3d2/ : FAILURE in 49m 32s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-multinode-full/89d228d/ : FAILURE in 1h 03m 08s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-nova-next-full/d29ac05/ : FAILURE in 46m 31s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-multinode-live-migration/7cbfbe2/ : FAILURE in 36m 42s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/dc6121b/ : FAILURE in 43m 02s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-full/e035abf/ : FAILURE in 57m 38s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron/ddfb75a/ : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":5},{"id":"fc69fc0e9649e9fa3cb44948965b2c290e89ae9b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-22 16:11:16.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/5/ : FAILURE in 2h 21m 16s","accounts_in_message":[],"_revision_number":5},{"id":"6e17bc0b17659d31553a94985ca13b3f48ae924e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-22 16:48:40.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2423c956060340f51a7f8e36760a7f6ee98d9efd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-22 17:28:45.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/5/check/tempest-dsvm-full-xenial/330e431/ : FAILURE in 39m 33s","accounts_in_message":[],"_revision_number":5},{"id":"3d0d6f75174dca516e78b0f2ff04779d32680d22","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-22 17:34:06.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/5/check/check-dsvm-tempest-vz7-exe-minimal/d834b53 : FAILURE in 44m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"417300b0be02498a4ccc03350b33be6c57c3cb6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-22 18:08:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/5/check/openstack-tox-pep8/d03741b/ : SUCCESS in 8m 35s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/5/check/openstack-tox-py27/6343ca6/ : POST_FAILURE in 5m 46s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/5/check/openstack-tox-py35/0272ad0/ : POST_FAILURE in 4m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/5/check/build-openstack-sphinx-docs/f519459/html/ : SUCCESS in 5m 51s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-py35/a0e01e0/ : FAILURE in 44m 25s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/5/check/build-openstack-releasenotes/ef9e409/html/ : SUCCESS in 8m 05s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional/88c1545/ : FAILURE in 5m 12s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional-py35/702ee9a/ : FAILURE in 5m 54s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode/d0f8bcd/ : SUCCESS in 1h 00m 50s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/daa08da/ : FAILURE in 1h 17m 38s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-cells/a6e62c5/ : FAILURE in 40m 29s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/eef469c/ : FAILURE in 53m 13s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-multinode-full/4492821/ : FAILURE in 58m 41s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-nova-next-full/b919829/ : FAILURE in 52m 15s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-multinode-live-migration/386f9b3/ : FAILURE in 35m 08s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c60f316/ : FAILURE in 36m 49s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-full/8321593/ : FAILURE in 44m 01s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron/c1dac39/ : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":5},{"id":"577e83e6d107a8fd2b6c14ac7d94f6f2ef8200bf","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-22 18:18:22.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3bb885883e81876303d263dab9574b1a22bb5cb7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-22 18:57:24.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/5/ : FAILURE in 2h 08m 15s","accounts_in_message":[],"_revision_number":5},{"id":"43c8c2a9edfe315624d7bcef55e5e8d154bc9b97","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-22 18:58:21.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/5/check/tempest-dsvm-full-xenial/dd51c8f/ : FAILURE in 39m 27s","accounts_in_message":[],"_revision_number":5},{"id":"acad6df7e08ba9ae07073cf1265aaf25a7b28b06","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-22 19:03:06.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/5/check/check-dsvm-tempest-vz7-exe-minimal/2f48bed : FAILURE in 44m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"d70097a798663bc3b530e0253b35f5ca99c422b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-22 19:31:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/5/check/openstack-tox-pep8/42a9255/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/5/check/openstack-tox-py27/ab85797/ : POST_FAILURE in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/5/check/openstack-tox-py35/345db8b/ : POST_FAILURE in 5m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/5/check/build-openstack-sphinx-docs/4c12a8a/html/ : SUCCESS in 5m 07s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-py35/d2fccfa/ : FAILURE in 52m 51s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/5/check/build-openstack-releasenotes/348fc21/html/ : SUCCESS in 8m 43s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional/b02977e/ : FAILURE in 4m 44s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional-py35/1f4a7bf/ : FAILURE in 4m 52s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode/f4991dd/ : SUCCESS in 59m 51s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/ffa32ed/ : FAILURE in 1h 11m 34s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-cells/60042cb/ : FAILURE in 36m 56s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/60d3f24/ : FAILURE in 54m 36s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-multinode-full/6d6ace9/ : FAILURE in 1h 04m 11s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-nova-next-full/b86c3ab/ : FAILURE in 45m 10s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-multinode-live-migration/fe4b948/ : FAILURE in 38m 07s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f7d12e5/ : FAILURE in 46m 57s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-full/159676b/ : FAILURE in 53m 55s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron/5fc2236/ : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":5},{"id":"dcb28f1d830d77c4a26b0cdf6f3ea0936a484d41","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-22 20:02:56.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b5f67ec117aecb6cd310d45a2becbb5b6e0fd9e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-22 20:48:11.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/5/check/tempest-dsvm-full-xenial/eb718cc/ : FAILURE in 44m 43s","accounts_in_message":[],"_revision_number":5},{"id":"ca92432185dc736906f6375c389cb26c97919900","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-22 20:51:24.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/5/check/check-dsvm-tempest-vz7-exe-minimal/1baf7b2 : FAILURE in 47m 53s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"b025e35aa8a601325f1867c111d922166803bed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-22 22:05:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/5/check/openstack-tox-pep8/0807e03/ : SUCCESS in 8m 49s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/5/check/openstack-tox-py27/94e8cdf/ : POST_FAILURE in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/5/check/openstack-tox-py35/4563dff/ : POST_FAILURE in 4m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/5/check/build-openstack-sphinx-docs/668a16d/html/ : SUCCESS in 6m 38s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-py35/925510a/ : FAILURE in 44m 49s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/5/check/build-openstack-releasenotes/9d91d1d/html/ : SUCCESS in 8m 06s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional/35ffd51/ : FAILURE in 9m 00s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/5/check/legacy-nova-tox-functional-py35/4e8f868/ : FAILURE in 4m 40s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode/6b16bd1/ : SUCCESS in 58m 28s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/14893c1/ : FAILURE in 1h 54m 43s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-cells/a72d23e/ : FAILURE in 38m 07s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3c47087/ : FAILURE in 1h 03m 08s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-multinode-full/7a80ddf/ : FAILURE in 1h 02m 43s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-nova-next-full/f804a42/ : FAILURE in 55m 13s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-multinode-live-migration/b9aa927/ : FAILURE in 36m 30s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ff17510/ : FAILURE in 37m 34s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/5/check/legacy-tempest-dsvm-neutron-full/911110b/ : FAILURE in 47m 06s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/5/check/legacy-grenade-dsvm-neutron/1e561c5/ : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3dc27980554c6ae547ebd56ff2be0cb9e182c49b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-22 22:27:19.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/499826/5/ : FAILURE in 2h 23m 59s","accounts_in_message":[],"_revision_number":5},{"id":"6114efaaccd15e2b14115d2f975040c7f5447923","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2017-10-23 09:46:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4d47d5d4cec86a9fb9a825e1c21804950305e69","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-23 11:55:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"84538565f341dda54c59bd3af5e98650f8683716","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-23 11:56:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"18c1ba5d8ed10d8959829ca6ee3468fc47387f8e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-23 12:44:44.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7461 : FAILURE in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f8715d7c80cf968b5d964b75de95242ad8e5d52","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-23 12:55:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/6/check/tempest-dsvm-full-xenial/597c481/ : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":6},{"id":"956df4b9b64bf66d1bcfc8fa708f24785812fc07","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 12:56:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":6},{"id":"765a32bbd017183b9b6dfff353c25f33584f29de","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-23 13:08:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/26/499826/6/check-nova/Nova-ML2-Sriov/f14b996 : SUCCESS in 1h 09m 00s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/26/499826/6/check-nova/Nova-MACVTAP-ML2-Sriov/7b70ddf : SUCCESS in 59m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"d67e28cd9322ee4a3de749ffaa7cea99adb41cc6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 13:09:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/6/check/check-dsvm-tempest-vz7-exe-minimal/0ef8a3f : SUCCESS in 1h 12m 53s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"ddcda5e2675304b8cd4740954e01d6ff65f592bf","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 13:27:02.000000000","message":"Patch Set 6: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/499826/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/051ab26 : SUCCESS in 1h 30m 17s","accounts_in_message":[],"_revision_number":6},{"id":"5731815c2c7c1ec27ccd3104f047aabd3028e5a2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 13:27:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":6},{"id":"eb69fe2f0b09b00f4afeff7b710cbd27c702a5f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-23 13:35:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/6/check/openstack-tox-pep8/ce9a0a1/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/6/check/openstack-tox-py27/90f6a47/ : SUCCESS in 15m 31s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/6/check/openstack-tox-py35/cf3f616/ : SUCCESS in 11m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/6/check/build-openstack-sphinx-docs/4875c2e/html/ : SUCCESS in 6m 12s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-py35/5e02a70/ : SUCCESS in 1h 18m 32s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/6/check/build-openstack-releasenotes/bb3f23d/html/ : SUCCESS in 11m 41s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/6/check/legacy-nova-tox-functional/5dfe680/ : FAILURE in 21m 06s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/6/check/legacy-nova-tox-functional-py35/272af79/ : FAILURE in 22m 56s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/6/check/legacy-grenade-dsvm-neutron-multinode/0d44495/ : SUCCESS in 1h 00m 30s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/6/check/legacy-grenade-dsvm-neutron-multinode-live-migration/86392b2/ : FAILURE in 1h 00m 24s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-cells/1b4bbf1/ : SUCCESS in 37m 06s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/481532d/ : SUCCESS in 1h 37m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-neutron-multinode-full/a06a389/ : FAILURE in 1h 30m 41s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-neutron-nova-next-full/6d9b050/ : SUCCESS in 1h 10m 04s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-multinode-live-migration/71dcc7d/ : SUCCESS in 42m 44s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/6/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/649658d/ : SUCCESS in 43m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/6/check/legacy-tempest-dsvm-neutron-full/ee68597/ : SUCCESS in 1h 28m 00s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/6/check/legacy-grenade-dsvm-neutron/f0aef05/ : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":6},{"id":"e351bf013458953b2a043cd9e4bf103ba813f80b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-23 13:46:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/499826/6/check/dsvm-tempest-neutron-network/a4e7e0d : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":6},{"id":"e1a9ce3681e95cc86b6f400155dac501ad747d4a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-23 14:23:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/499826/6/ : SUCCESS in 2h 26m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a9c6030f15eb57375c2cc931af47ba8ee156eaf2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-23 17:27:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)\n\nThere\u0027s microversion confusion in the gabbi files: unless it is explicitly stated some of the file are minimum microversion.","accounts_in_message":[],"_revision_number":6},{"id":"5fa6663329a24c4597d1e23948ef8108833ed7b9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-23 19:03:59.000000000","message":"Patch Set 6:\n\n(7 comments)\n\nThanks for the help, Chris.","accounts_in_message":[],"_revision_number":6},{"id":"0bcdf2bbedc8e207540c6c3b8ebadb491c819dc2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-23 19:08:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1c5f22db29685ba1f2c139e7d653000ea5987833","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-23 19:58:03.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7490 : FAILURE in 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b6e189e11cc20601fba90e95a263439ae32f346e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-23 20:04:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/26/499826/7/check/tempest-dsvm-full-xenial/20f7b21/ : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":7},{"id":"5024d735de680864c2e267bdf79bade71b0fd22b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 20:04:35.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"ce63c479ebbec417e778c099c8d3b4e3298217bc","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 20:10:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/26/499826/7/check/check-dsvm-tempest-vz7-exe-minimal/23e7ea0 : SUCCESS in 1h 01m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"3caaea533eae7ba8da8bf61c8220b00abdca2086","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-23 20:15:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/26/499826/7/check/nova-out-of-tree-pvm/15ef065 : SUCCESS in 1h 06m 23s\n- nova-in-tree-pvm http://184.172.12.213/26/499826/7/check/nova-in-tree-pvm/2a345a5 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":7},{"id":"0fc00ff34ed630f65e0832b9b41dcfc308b1c0fc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 20:18:59.000000000","message":"Patch Set 7: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/499826/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e192027 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":7},{"id":"ee0af668ff28a5a60c66d70a813a2418ddae9e87","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 20:19:23.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"87e12c91b88082dbd67a7d14a20665a3a0005325","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-23 21:09:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/26/499826/7/check/dsvm-tempest-neutron-network/12753a4 : SUCCESS in 2h 01m 00s","accounts_in_message":[],"_revision_number":7},{"id":"451be7c74b0c72daccd6dbb52e8b01d6fcbe9b7a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-23 21:45:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/499826/7/ : SUCCESS in 2h 10m 51s","accounts_in_message":[],"_revision_number":7},{"id":"df2f603d5d685a583ec35072ff0e2c19d16bfb0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-23 22:10:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/7/check/openstack-tox-pep8/3f225e5/ : SUCCESS in 9m 49s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/7/check/openstack-tox-py27/b56eb06/ : SUCCESS in 13m 10s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/7/check/openstack-tox-py35/b24f2b7/ : SUCCESS in 13m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/7/check/build-openstack-sphinx-docs/85f5bcf/html/ : SUCCESS in 6m 17s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-py35/7c2d3e2/ : SUCCESS in 1h 23m 48s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/7/check/build-openstack-releasenotes/1ffe2d1/html/ : SUCCESS in 8m 34s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/7/check/legacy-nova-tox-functional/76f9441/ : SUCCESS in 30m 54s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/7/check/legacy-nova-tox-functional-py35/f19e0b5/ : SUCCESS in 22m 00s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/26/499826/7/check/legacy-grenade-dsvm-neutron-multinode/7896685/ : SUCCESS in 1h 05m 01s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/26/499826/7/check/legacy-grenade-dsvm-neutron-multinode-live-migration/c2a5de6/ : FAILURE in 59m 44s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-cells/985e05d/ : SUCCESS in 46m 10s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/84992fc/ : SUCCESS in 1h 14m 21s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-neutron-multinode-full/18e3b2e/ : SUCCESS in 1h 44m 57s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-neutron-nova-next-full/267ff13/ : SUCCESS in 1h 17m 28s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-multinode-live-migration/18b61b1/ : SUCCESS in 55m 12s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/26/499826/7/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/dfe722f/ : SUCCESS in 50m 14s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/7/check/legacy-tempest-dsvm-neutron-full/1c75336/ : SUCCESS in 1h 41m 13s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/7/check/legacy-grenade-dsvm-neutron/b91df76/ : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":7},{"id":"f8fc8d637b11194df893174d0520668474ce8003","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-23 22:10:28.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"1c94243dec6baa5b9556784cc45f3c3d7dcb5cdf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-23 23:18:16.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-26-499826-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"5f0d79b1020ea1e49050f50973de14dbc074be8b","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-24 02:02:40.000000000","message":"Patch Set 7: Code-Review+1\n\nLGTM, but seems so many 1.11 on the fly..","accounts_in_message":[],"_revision_number":7},{"id":"ea43b6c86a38f52a8ac10c79d9a5752cd9c5b5fa","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-10-25 16:35:56.000000000","message":"Patch Set 7: Code-Review+2\n\nYeah, there is a lot of candidates for v1.11.\nI feel this is mature and simple than the other ones at this time.","accounts_in_message":[],"_revision_number":7},{"id":"736bf4ac43234d384f83ec96a4fc51555c5fd7bd","author":{"_account_id":1063,"name":"Ed Leafe","email":"ed@leafe.com","username":"ed-leafe"},"date":"2017-10-25 18:46:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"6b53b0071e69e541bfefa6e13ce83a14865dd557","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2017-10-27 15:07:07.000000000","message":"Patch Set 7: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"20f4ff27bdf303faddc16e3858f1a9dd2c22514c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 15:11:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5b09d4ccc04135c884a10fad1db104f122f0537d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 17:45:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/26/499826/7/gate/openstack-tox-pep8/1f1dc0e/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/26/499826/7/gate/openstack-tox-py27/a2d976c/ : SUCCESS in 12m 53s\n- openstack-tox-py35 http://logs.openstack.org/26/499826/7/gate/openstack-tox-py35/6fa234b/ : SUCCESS in 13m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/499826/7/gate/build-openstack-sphinx-docs/fb83c7f/html/ : SUCCESS in 6m 23s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/26/499826/7/gate/legacy-tempest-dsvm-py35/86ed61f/ : SUCCESS in 2h 28m 11s\n- build-openstack-releasenotes http://logs.openstack.org/26/499826/7/gate/build-openstack-releasenotes/3b92000/html/ : SUCCESS in 7m 05s\n- legacy-nova-tox-functional http://logs.openstack.org/26/499826/7/gate/legacy-nova-tox-functional/83ce11d/ : SUCCESS in 24m 04s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/26/499826/7/gate/legacy-nova-tox-functional-py35/02b8040/ : SUCCESS in 17m 57s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/26/499826/7/gate/legacy-tempest-dsvm-cells/f19d389/ : SUCCESS in 43m 29s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/26/499826/7/gate/legacy-tempest-dsvm-multinode-live-migration/360c34f/ : SUCCESS in 44m 39s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/26/499826/7/gate/legacy-tempest-dsvm-neutron-full/6938d25/ : SUCCESS in 1h 20m 35s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/26/499826/7/gate/legacy-grenade-dsvm-neutron/246025f/ : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":7},{"id":"85fa33dc189d2c9f832aa92f14d8609ddf60a6e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-27 17:45:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a9105b4904b23f640537b36b6b474e74dab26541","revisions":{"1e7880e696633ef9ba7d45ae93655da11da060f1":{"kind":"REWORK","_number":1,"created":"2017-08-31 22:07:28.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/1"}}},"commit":{"parents":[{"commit":"2a4ca8bd6aa40ccd26300feaef4267aa71f69abf","subject":"Merge \"Fix _delete_inventory log message in report client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a4ca8bd6aa40ccd26300feaef4267aa71f69abf"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link for all\nmicroversions.\n\n[1] https://review.openstack.org/#/c/366789/\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e7880e696633ef9ba7d45ae93655da11da060f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e7880e696633ef9ba7d45ae93655da11da060f1"}]},"branch":"refs/heads/master"},"692816355ff28b9f9d6b4991388cb5ac7198de0c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-09-01 13:23:49.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/2"}}},"commit":{"parents":[{"commit":"2a4ca8bd6aa40ccd26300feaef4267aa71f69abf","subject":"Merge \"Fix _delete_inventory log message in report client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a4ca8bd6aa40ccd26300feaef4267aa71f69abf"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-09-01 13:23:44.000000000","tz":0},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link for all\nmicroversions.\n\n[1] https://review.openstack.org/#/c/366789/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/692816355ff28b9f9d6b4991388cb5ac7198de0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/692816355ff28b9f9d6b4991388cb5ac7198de0c"}]},"branch":"refs/heads/master"},"b27abbe2c67d13ee959203f9742668f4f46dfe3e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2017-09-01 14:21:19.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/3"}}},"commit":{"parents":[{"commit":"2a4ca8bd6aa40ccd26300feaef4267aa71f69abf","subject":"Merge \"Fix _delete_inventory log message in report client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a4ca8bd6aa40ccd26300feaef4267aa71f69abf"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-09-01 14:21:15.000000000","tz":0},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link for all\nmicroversions.\n\nStrictly speaking, this is an API change worthy of a new microversion\nitself.  However, it\u0027s eminently compatible; and the change which\nsplit which links show up depending on the microversion [2] broke that\nrule similarly.\n\n[1] https://review.openstack.org/#/c/366789/\n[2] https://review.openstack.org/#/c/468923/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b27abbe2c67d13ee959203f9742668f4f46dfe3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b27abbe2c67d13ee959203f9742668f4f46dfe3e"}]},"branch":"refs/heads/master"},"0bf951fce49e2ec17137e4dfe4f65078fd1b9752":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-10-20 20:15:58.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/4"}}},"commit":{"parents":[{"commit":"9058ed951f58d78f5de7ea40a7e305170dd3c005","subject":"Merge \"Move lvm handling to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9058ed951f58d78f5de7ea40a7e305170dd3c005"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-20 20:15:58.000000000","tz":0},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link for all\nmicroversions.\n\nStrictly speaking, this is an API change worthy of a new microversion\nitself.  However, it\u0027s eminently compatible; and the change which\nsplit which links show up depending on the microversion [2] broke that\nrule similarly.\n\n[1] https://review.openstack.org/#/c/366789/\n[2] https://review.openstack.org/#/c/468923/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0bf951fce49e2ec17137e4dfe4f65078fd1b9752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0bf951fce49e2ec17137e4dfe4f65078fd1b9752"}]},"branch":"refs/heads/master"},"88e9c9390be6ea8798c7445f867c059653ba05e2":{"kind":"REWORK","_number":5,"created":"2017-10-20 20:36:54.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/5"}}},"commit":{"parents":[{"commit":"9058ed951f58d78f5de7ea40a7e305170dd3c005","subject":"Merge \"Move lvm handling to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9058ed951f58d78f5de7ea40a7e305170dd3c005"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-20 20:36:27.000000000","tz":-300},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link in a new\nmicroversion, 1.11\n\n[1] https://review.openstack.org/#/c/366789/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88e9c9390be6ea8798c7445f867c059653ba05e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88e9c9390be6ea8798c7445f867c059653ba05e2"}]},"branch":"refs/heads/master"},"d4b42c54872c53562a057f891f466d36aea8fd01":{"kind":"REWORK","_number":6,"created":"2017-10-23 11:55:58.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/6"}}},"commit":{"parents":[{"commit":"9058ed951f58d78f5de7ea40a7e305170dd3c005","subject":"Merge \"Move lvm handling to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9058ed951f58d78f5de7ea40a7e305170dd3c005"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-23 11:55:55.000000000","tz":0},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link in a new\nmicroversion, 1.11\n\n[1] https://review.openstack.org/#/c/366789/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4b42c54872c53562a057f891f466d36aea8fd01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4b42c54872c53562a057f891f466d36aea8fd01"}]},"branch":"refs/heads/master"},"a9105b4904b23f640537b36b6b474e74dab26541":{"kind":"REWORK","_number":7,"created":"2017-10-23 19:08:05.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/26/499826/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/499826/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/499826/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/499826/7"}}},"commit":{"parents":[{"commit":"9058ed951f58d78f5de7ea40a7e305170dd3c005","subject":"Merge \"Move lvm handling to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9058ed951f58d78f5de7ea40a7e305170dd3c005"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-08-31 22:02:33.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-23 19:08:05.000000000","tz":-300},"subject":"Include /resource_providers/uuid/allocations link","message":"Include /resource_providers/uuid/allocations link\n\n/resource_providers/{rp_uuid}/allocations has been available since\nmicroversion 1.0 [1], but wasn\u0027t listed in the \"links\" section of the\nGET /resource_providers response.  This change adds the link in a new\nmicroversion, 1.11\n\n[1] https://review.openstack.org/#/c/366789/\n\nCloses-Bug: #1714275\n\nChange-Id: I6a1d320ce914926791d5f45e89bf4c601a6b10a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9105b4904b23f640537b36b6b474e74dab26541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9105b4904b23f640537b36b6b474e74dab26541"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
