)]}'
{"id":"openstack%2Fnova~502056","triplet_id":"openstack%2Fnova~master~Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a","project":"openstack/nova","branch":"master","topic":"bug/1715437","hashtags":[],"change_id":"Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a","subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","status":"MERGED","created":"2017-09-08 13:20:00.000000000","updated":"2017-10-05 10:03:23.000000000","submitted":"2017-10-05 10:03:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"502056-1507197803701-51935e5d","meta_rev_id":"ffb8addd8c0a7a9e587d4f2697e316febbafa9e7","_number":502056,"virtual_id_number":502056,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2017-10-04 10:26:50.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-04 10:34:51.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2017-10-05 10:03:23.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2017-10-04 09:00:30.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":-1,"date":"2017-10-04 11:08:42.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2017-10-04 09:50:53.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2017-10-04 18:17:23.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2017-10-05 09:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-05 09:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-16 08:21:24.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2017-09-16 10:41:38.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-21 07:03:45.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2017-09-21 08:09:27.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2017-10-03 16:18:51.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-10-04 09:00:30.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 09:50:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 10:26:50.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 10:34:51.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 11:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-04 18:17:23.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2017-10-05 09:26:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2017-10-05 10:03:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5f5861988a68790461b559ac3d8c86ce05e9a28e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-08 13:20:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbc221423fb39189ee0b92d5f3558a31b0d965e6","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-08 15:43:03.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/56/502056/1/check/nova-out-of-tree-pvm/48cfaf0 : SUCCESS in 2h 21m 04s\n- nova-in-tree-pvm http://184.172.12.213/56/502056/1/check/nova-in-tree-pvm/83a0854 : FAILURE in 40m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fdad55e370888ca5d58e6c196f19454e9b9d68c6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-08 15:51:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/502056/1/ : SUCCESS in 2h 12m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f205c240965feaabbd2af420360711e9b5201214","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-08 17:01:38.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/502056/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9b61814 : FAILURE in 3h 36m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f4433de7fe21334c299e9b8cab3c8f0f37a14c68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-08 22:25:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/502056/1/check/gate-nova-docs-ubuntu-xenial/354e6f4//doc/build/html/ : SUCCESS in 5m 15s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/502056/1/check/gate-nova-pep8-ubuntu-xenial/1fb6c1a/ : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d3381bdeecf185b705bd69618ce77d715ed71e5d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-11 21:22:21.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/502056/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a739ea4ad38f1d30707c39dd13fd49b1d500cc5f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2017-09-12 08:46:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a135b0fc3e78e594f858a61a70a4c2dca083c3d1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-15 16:35:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c9a5fd9623f87f60a70f25b01bcfd705b889ce84","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-15 16:41:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b061d6902152c45e8fec29f71c773ccc46f3b1b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-15 19:32:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/502056/2/check/gate-nova-docs-ubuntu-xenial/4440837//doc/build/html/ : SUCCESS in 5m 19s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/502056/2/check/gate-nova-pep8-ubuntu-xenial/c2a93c0/ : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c7131dfa9b10c7152eb86f5ebafaacffe203c609","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-15 20:45:46.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-56-502056-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"40e91fb0580f8ef180eb61332e780aea34e98909","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-15 23:16:20.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/56/502056/2/check/nova-out-of-tree-pvm/dd4c133 : FAILURE in 3h 57m 13s\n- nova-in-tree-pvm http://184.172.12.213/56/502056/2/check/nova-in-tree-pvm/6863a21 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b1356467e94f14ac556a6f41cfa51c5f4c8de1cf","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-16 00:08:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/502056/2/ : SUCCESS in 2h 21m 28s","accounts_in_message":[],"_revision_number":2},{"id":"13f99a9a4a77b19e8e547c5be65d80495cad626b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-16 02:39:12.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/502056/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/3824419 : SUCCESS in 3h 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6bb5a9304855be4a3c824dc6c2805a57b91c5c2e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2017-09-16 08:21:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c598f7f7ee27f75e97fec14445b01ddc9ed4d96","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2017-09-16 10:41:38.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/502056/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2404f23c0634ed96dfafe639a9eff775ba8cd8f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-18 14:53:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"453ed9e03c66c5ed59cc78a4e95564e5e203a0c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-19 13:30:16.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a2fa07d4fd63b67b536bc8d9d643721b0582bd11","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-09-20 13:49:34.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"72acc7380da79a8c88747acdc0df817d71aebce9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 14:10:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/502056/3/check/gate-nova-docs-ubuntu-xenial/39db277//doc/build/html/ : SUCCESS in 5m 06s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/502056/3/check/gate-nova-pep8-ubuntu-xenial/1631846/ : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":3},{"id":"54831a2966862a3f46add1e9cfedec7e94295fec","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-09-20 14:15:18.000000000","message":"Patch Set 3: Code-Review+2\n\nI think that this is a step in the good direction and additional information about how to configure nova for the RT use case could be done later in a separate patch.","accounts_in_message":[],"_revision_number":3},{"id":"151d80a81a6cbad8b91f259292a4e6530c868981","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-09-20 15:58:03.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-56-502056-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"c8a7e7fda93486917d1fd74cfb4ecd40f68ba93a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-09-20 16:04:48.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/56/502056/3/check/nova-out-of-tree-pvm/039f0d3 : SUCCESS in 2h 12m 18s\n- nova-in-tree-pvm http://184.172.12.213/56/502056/3/check/nova-in-tree-pvm/94a9d58 : FAILURE in 2h 10m 38s","accounts_in_message":[],"_revision_number":3},{"id":"92cd70ceb43985819a14678b36bad77802f9d20e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-09-20 17:04:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/502056/3/ : SUCCESS in 2h 16m 35s","accounts_in_message":[],"_revision_number":3},{"id":"96a418866701214bb175399347561b665a7427a0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-09-21 03:39:19.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/502056/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/388e7b5 : SUCCESS in 2h 58m 02s","accounts_in_message":[],"_revision_number":3},{"id":"6db77884daf3adb64dc78c8910ee0a92d9352531","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2017-09-21 07:03:45.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b626c84d323955f657a861cc1a871e88d1d630f4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2017-09-21 08:09:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5100c7a3f20d3287618f7a9ca595ee5d63dd564b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-03 10:31:59.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1478fe053fc0dcd5d925ea7827aa20f957ed234a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-03 10:45:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5551c4f78e4b26103aab8a8f451275106be0b15d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-03 11:17:54.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM! I think it is OK to say that the core mask should start with ^ even if nova today allows more than that. \n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":4},{"id":"d77e60c6c7e63e73e22aa547c004cac16174f4f3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-03 11:24:38.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/56/502056/4/check/nova-out-of-tree-pvm/1ad5ae4 : FAILURE in 20m 49s\n- nova-in-tree-pvm http://184.172.12.213/56/502056/4/check/nova-in-tree-pvm/eec6d4e : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":4},{"id":"73df3cfcc08fb5cee12ebeb9deff900312f0733a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-03 12:38:15.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/502056/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1482225 : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":4},{"id":"d728817b5180c43696653989c754e5d3f5d7a432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-03 14:10:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/502056/4/check/build-openstack-sphinx-docs/89fb4d1/html/latest/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/56/502056/4/check/openstack-tox-pep8/9402ffd/ : SUCCESS in 11m 38s\n- legacy-tempest-dsvm-py35 legacy-tempest-dsvm-py35 : NODE_FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"fdfd6113aed5d36844706efdcf6b315e13ab6bb0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-03 15:21:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nA request inline to provide some documentation (or link) to what \"correctly configured host\" happens to mean.\n\nBest,\n\n-jay","accounts_in_message":[],"_revision_number":4},{"id":"a0a3578d55cad11837ae4a40360ee207a6abc155","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-03 16:18:51.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/502056/4/ : FAILURE in 3h 50m 02s","accounts_in_message":[],"_revision_number":4},{"id":"350a09033fe2ea0362ff8ca9531d1c0ddca31017","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-04 08:37:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8fd73ace113e1bc6030d62df89f054c9af52089f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-10-04 08:47:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"66903a4806b3cdcbb2834af7cf8a32912da8a3a2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-04 09:00:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/502056/5/check/gate-nova-docs-ubuntu-xenial/feecfd5//doc/build/html/ : SUCCESS in 4m 42s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/502056/5/check/gate-nova-pep8-ubuntu-xenial/c7ddfe9/ : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":5},{"id":"376b834c4221bc07c5bc7017c7ec41a60d61c8e4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-04 09:00:30.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"648259a20d5a796e98eb5d388fd9e5798fcdeab3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-04 09:50:53.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-56-502056-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"f05886459981b214d64d675a85adfaae7e43440b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-04 10:26:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/56/502056/5/check/nova-out-of-tree-pvm/35a0d2f : SUCCESS in 1h 38m 25s\n- nova-in-tree-pvm http://184.172.12.213/56/502056/5/check/nova-in-tree-pvm/16fce51 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":5},{"id":"3b9b673a519e52b913eb014fd564b7891bc6d899","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-04 10:34:51.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/502056/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/112d7cb : SUCCESS in 1h 46m 05s","accounts_in_message":[],"_revision_number":5},{"id":"ce494ae55cbcecc6c089ebe5da4afc8382125d85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-04 11:08:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/56/502056/5/check/build-openstack-sphinx-docs/6799d01/html/latest/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/56/502056/5/check/openstack-tox-pep8/7bcd4bf/ : SUCCESS in 7m 18s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/56/502056/5/check/legacy-tempest-dsvm-py35/5a90ed5/ : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3ef10a0b6a13a9fc9a0d9c4a82a7708d5f81caf1","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-10-04 18:17:23.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks for adding that TODO. ++ from me now.","accounts_in_message":[],"_revision_number":5},{"id":"61b569b6fcd0fb400b675870cf219f42f45713cc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-10-05 09:26:59.000000000","message":"Patch Set 5: Workflow+1 Code-Review+2\n\nLet\u0027s get this merged. Further improvements can be done in follow ups.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":5},{"id":"7dda47c25f8a2a0aa59481c7b25074e98d6ff2f5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 09:27:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"0e9e3b4477203a67b5a6c5839504ed22dd6524d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 10:03:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/56/502056/5/gate/gate-nova-docs-ubuntu-xenial/4f6fd3a//doc/build/html/ : SUCCESS in 5m 19s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/56/502056/5/gate/gate-nova-pep8-ubuntu-xenial/4144543/ : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":5},{"id":"9f8460e06608f6538a3f51fa72f40fc28f53fb64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 10:03:23.000000000","message":"Change has been successfully merged by Jenkins","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bd3a4d242f7b0208d7ddb2571b31975ced91946e","revisions":{"f52da207dcf5d408b017104fd158cf0cff293865":{"kind":"REWORK","_number":1,"created":"2017-09-08 13:20:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/502056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/502056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/502056/1"}}},"commit":{"parents":[{"commit":"62135e7b00bf5b620ce62bb7f297b2a3409bcd42","subject":"doc: Add documentation for emulator_thread_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62135e7b00bf5b620ce62bb7f297b2a3409bcd42"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:05:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:19:25.000000000","tz":60},"subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","message":"doc: Add documentation for cpu_realtime, cpu_realtime_mask\n\nThis wasn\u0027t documented anywhere but the spec [1]. Fix this.\n\nWe may want to provide a more in-depth overview of using RT features of\nOpenStack, but that\u0027s a future work item.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/mitaka/implemented/libvirt-real-time.html\n\nChange-Id: Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a\nCloses-Bug: #1715437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f52da207dcf5d408b017104fd158cf0cff293865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f52da207dcf5d408b017104fd158cf0cff293865"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"62135e7b00bf5b620ce62bb7f297b2a3409bcd42","is_merged_in_target_branch":false,"change_id":"I8d79e1f1277fea9a070790a99bebcefe22bed433","change_number":501721,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7843637f136a7e1d444c6fd242f5a4b5b3cc5b13":{"kind":"REWORK","_number":2,"created":"2017-09-15 16:41:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/502056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/502056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/502056/2"}}},"commit":{"parents":[{"commit":"cea4b1ae669b6f76044d54e7c08734b181e8b469","subject":"doc: Add documentation for emulator_thread_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cea4b1ae669b6f76044d54e7c08734b181e8b469"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:05:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-15 16:40:51.000000000","tz":-360},"subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","message":"doc: Add documentation for cpu_realtime, cpu_realtime_mask\n\nThis wasn\u0027t documented anywhere but the spec [1]. Fix this.\n\nWe may want to provide a more in-depth overview of using RT features of\nOpenStack, but that\u0027s a future work item.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/mitaka/implemented/libvirt-real-time.html\n\nChange-Id: Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a\nCloses-Bug: #1715437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7843637f136a7e1d444c6fd242f5a4b5b3cc5b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7843637f136a7e1d444c6fd242f5a4b5b3cc5b13"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cea4b1ae669b6f76044d54e7c08734b181e8b469","is_merged_in_target_branch":false,"change_id":"I8d79e1f1277fea9a070790a99bebcefe22bed433","change_number":501721,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5fd26bbba5dbe91a5a4387baf99fdd78c82c1f30":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-09-20 13:49:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/502056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/502056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/502056/3"}}},"commit":{"parents":[{"commit":"0b77e3cdd820ec1039bcea9d3a0431d98568f8b1","subject":"doc: Add documentation for emulator_thread_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b77e3cdd820ec1039bcea9d3a0431d98568f8b1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:05:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-20 13:06:43.000000000","tz":60},"subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","message":"doc: Add documentation for cpu_realtime, cpu_realtime_mask\n\nThis wasn\u0027t documented anywhere but the spec [1]. Fix this.\n\nWe may want to provide a more in-depth overview of using RT features of\nOpenStack, but that\u0027s a future work item.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/mitaka/implemented/libvirt-real-time.html\n\nChange-Id: Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a\nCloses-Bug: #1715437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fd26bbba5dbe91a5a4387baf99fdd78c82c1f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fd26bbba5dbe91a5a4387baf99fdd78c82c1f30"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0b77e3cdd820ec1039bcea9d3a0431d98568f8b1","is_merged_in_target_branch":true,"change_id":"I8d79e1f1277fea9a070790a99bebcefe22bed433","change_number":501721,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4be354f533544e234b703bc778c5fd2f4631662f":{"kind":"REWORK","_number":4,"created":"2017-10-03 10:45:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/502056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/502056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/502056/4"}}},"commit":{"parents":[{"commit":"cfff910b0d1a0d9f24b6c1596ceef8dd6b8b3ac6","subject":"Merge \"Do not setup conductor in BaseAPITestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfff910b0d1a0d9f24b6c1596ceef8dd6b8b3ac6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:05:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-10-03 10:31:33.000000000","tz":60},"subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","message":"doc: Add documentation for cpu_realtime, cpu_realtime_mask\n\nThis wasn\u0027t documented anywhere but the spec [1]. Fix this.\n\nWe may want to provide a more in-depth overview of using RT features of\nOpenStack, but that\u0027s a future work item.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/mitaka/implemented/libvirt-real-time.html\n\nChange-Id: Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a\nCloses-Bug: #1715437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4be354f533544e234b703bc778c5fd2f4631662f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4be354f533544e234b703bc778c5fd2f4631662f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cfff910b0d1a0d9f24b6c1596ceef8dd6b8b3ac6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bd3a4d242f7b0208d7ddb2571b31975ced91946e":{"kind":"REWORK","_number":5,"created":"2017-10-04 08:47:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/502056/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/56/502056/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/56/502056/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/56/502056/5"}}},"commit":{"parents":[{"commit":"7ab5ee982da5a9122359437390822f6839367198","subject":"Merge \"Move allocation manipulation out of drop_move_claim()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ab5ee982da5a9122359437390822f6839367198"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-09-08 13:05:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2017-10-04 08:43:22.000000000","tz":60},"subject":"doc: Add documentation for cpu_realtime, cpu_realtime_mask","message":"doc: Add documentation for cpu_realtime, cpu_realtime_mask\n\nThis wasn\u0027t documented anywhere but the spec [1]. Fix this.\n\nWe may want to provide a more in-depth overview of using RT features of\nOpenStack, but that\u0027s a future work item.\n\n[1] https://specs.openstack.org/openstack/nova-specs/specs/mitaka/implemented/libvirt-real-time.html\n\nChange-Id: Id30bc8447a6b482ad114ec6ebd3d5dab20ca0e3a\nCloses-Bug: #1715437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd3a4d242f7b0208d7ddb2571b31975ced91946e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd3a4d242f7b0208d7ddb2571b31975ced91946e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ab5ee982da5a9122359437390822f6839367198","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
