)]}'
{"id":"openstack%2Fnova~511715","triplet_id":"openstack%2Fnova~master~I2a19d8fe8dda5754112361be5ceac8a0dc747ae9","project":"openstack/nova","branch":"master","topic":"bp/placement-doc-enhancement-queens","hashtags":[],"change_id":"I2a19d8fe8dda5754112361be5ceac8a0dc747ae9","subject":"[placement]Enhance doc for placement allocation list","status":"MERGED","created":"2017-10-13 05:48:05.000000000","updated":"2017-12-05 13:32:37.000000000","submitted":"2017-12-05 13:32:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"511715-1512480757810-d6052c5b","meta_rev_id":"e42901fd31424b2e09b3d49f176c0f2fd6627101","_number":511715,"virtual_id_number":511715,"owner":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-11-30 22:06:21.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-11-30 22:25:14.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2017-12-01 00:59:46.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-11-30 21:35:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2017-12-05 13:32:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2017-12-01 09:51:16.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2017-11-30 22:48:54.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2017-12-05 11:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-12-01 06:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2017-11-30 21:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":2,"date":"2017-11-30 20:41:26.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2017-12-05 11:37:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-13 09:01:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-10-31 09:26:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-31 10:44:18.000000000","updated_by":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"reviewer":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"state":"REVIEWER"},{"updated":"2017-10-31 12:16:52.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-11-27 17:44:29.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2017-11-29 10:18:51.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-29 12:49:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-30 20:41:26.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2017-11-30 21:03:05.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2017-11-30 21:35:15.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-30 22:06:21.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-30 22:25:14.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-30 22:48:54.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-01 00:59:46.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-01 06:00:06.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2017-12-01 09:51:16.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-12-05 11:37:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2017-12-05 13:32:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f1815526412d907338fc485eb5f902f65dbc624","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-13 05:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e569bde9f70820cfae72bd67b1823c3d2610c30","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-13 05:50:03.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/1/check/check-dsvm-tempest-vz7-exe-minimal/ebc081c : FAILURE in 1m 04s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"655eeda4ac0a36871a4213faf8819194d93d1013","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-13 05:51:07.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/15/511715/1/check/dsvm-tempest-xen/a293fe2 : FAILURE in 40s","accounts_in_message":[],"_revision_number":1},{"id":"1531bfa3b29d2dd68485c0faaaf10d52a2894e5d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-13 06:48:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/1/check/tempest-dsvm-full-xenial/52cefdb/ : SUCCESS in 59m 21s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/1/check/grenade-dsvm-xenial/54ae1f1/ : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9907f82f4dde8185cb32ac8ce2ba4693035d1b4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-13 06:48:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"cb9727453f48e9592bf4218c1ded30b02408de97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 07:00:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/15/511715/1/check/gate-nova-docs-ubuntu-xenial/68a555c//doc/build/html/ : SUCCESS in 4m 20s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/15/511715/1/check/gate-nova-pep8-ubuntu-xenial/069d9b5/ : SUCCESS in 8m 34s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/15/511715/1/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/f4ad421/ : FAILURE in 1h 08m 40s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/15/511715/1/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6cd45fc/ : SUCCESS in 39m 48s\n- gate-placement-api-ref http://logs.openstack.org/15/511715/1/check/gate-placement-api-ref/249f424/ : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8417d81642e15719294a4fe3914906e0dc614e94","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-13 07:39:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/1/check/nova-out-of-tree-pvm/964c91e : SUCCESS in 1h 47m 59s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/1/check/nova-in-tree-pvm/9a75028 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e941369745ba34e3b1bab3e50cf63d6ddb7bd8d2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-13 07:48:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/1/check/dsvm-tempest-neutron-network/9d3cb0b : SUCCESS in 1h 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0197bfed9c43a24495836e1b757d2c903b8d21cb","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-13 07:53:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8a784342274047da10e4e2f3c8d08f42d690761","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-13 07:54:29.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/15/511715/2/check/dsvm-tempest-xen/5ec46f6 : FAILURE in 33s","accounts_in_message":[],"_revision_number":2},{"id":"b9b1e9abf17e893dd70e450a93e08297ea8e0015","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-13 08:46:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/2/check/tempest-dsvm-full-xenial/c519140/ : SUCCESS in 53m 02s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/2/check/grenade-dsvm-xenial/22aca71/ : SUCCESS in 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ddf9c64df718f384558a9bf4cbc7be6a73d63d6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-13 08:47:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"6a10cf314a8e469357e81eae351081f8d9084480","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-13 08:55:39.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/2/check/check-dsvm-tempest-vz7-exe-minimal/d700d47 : FAILURE in 1h 01m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"97cba3bd284d93dacf903d9f996969e4fa92f9b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 09:01:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-docs-ubuntu-xenial http://docs-draft.openstack.org/15/511715/2/check/gate-nova-docs-ubuntu-xenial/0908b7c//doc/build/html/ : SUCCESS in 4m 31s\n- gate-nova-pep8-ubuntu-xenial http://logs.openstack.org/15/511715/2/check/gate-nova-pep8-ubuntu-xenial/9e679ba/ : SUCCESS in 9m 04s\n- gate-grenade-dsvm-neutron-multinode-live-migration-nv http://logs.openstack.org/15/511715/2/check/gate-grenade-dsvm-neutron-multinode-live-migration-nv/2fe36da/ : FAILURE in 1h 02m 34s (non-voting)\n- gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial http://logs.openstack.org/15/511715/2/check/gate-tempest-dsvm-multinode-live-migration-ubuntu-xenial/6f0e1f0/ : SUCCESS in 39m 00s\n- gate-placement-api-ref http://docs-draft.openstack.org/15/511715/2/check/gate-placement-api-ref/7758af7//placement-api-ref/build/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b31dd64ac74f81f524e546d653aea49d383c1956","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-13 09:01:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"7db54c61591f4e156b82cba9b5c81d2acc30feb9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-13 09:35:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/2/check/nova-out-of-tree-pvm/9a36abb : SUCCESS in 1h 00m 47s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/2/check/nova-in-tree-pvm/a749dc5 : SUCCESS in 1h 42m 20s","accounts_in_message":[],"_revision_number":2},{"id":"97e01eae2475cee86adc5965ac4b6413342e7b51","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-13 09:45:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/2/check/dsvm-tempest-neutron-network/8226c1d : SUCCESS in 1h 51m 44s","accounts_in_message":[],"_revision_number":2},{"id":"afc5a1aa0f84ae9c639019e98d1acfded6d5f09a","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-13 15:45:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nSee within, both of these changes are not correct.","accounts_in_message":[],"_revision_number":2},{"id":"de537cb93aa42b4af3469ad87373cc48ceeebe5a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-13 16:30:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/511715/2/ : FAILURE in 3h 59m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6a7336ef050e380be430e5c485c8b7a7d9cf489f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-16 06:57:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/15/511715/2/check-nova/Nova-ML2-Sriov/d64e2c9 : SUCCESS in 57m 00s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/15/511715/2/check-nova/Nova-MACVTAP-ML2-Sriov/add9408 : SUCCESS in 52m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"9880881bf284bc8f5222d394a5f93dbde1e6848b","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-18 03:14:00.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8605eb228b3c33eaae82a6663c1c68ed663fc398","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-19 01:59:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e26d561e71c3f6e5b5ce64838cae418792c76bbe","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 02:41:40.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/3/check/nova-out-of-tree-pvm/160b479 : FAILURE in 39m 45s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/3/check/nova-in-tree-pvm/1425edb : FAILURE in 41m 29s","accounts_in_message":[],"_revision_number":3},{"id":"802641713c84eb634204f34b3ac8aa692e73434c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-19 02:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/3/check/tempest-dsvm-full-xenial/ad495fb/ : SUCCESS in 58m 32s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/3/check/grenade-dsvm-xenial/eb60bb7/ : FAILURE in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d91df8d11b120f4773971097748661490d160bc7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 02:59:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/3/check/check-dsvm-tempest-vz7-exe-minimal/1e3bf3a : SUCCESS in 58m 59s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"2c5bd60359cef0b9a1a310b531298266c733b8dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 02:59:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"cb8117b0ca8bc5b36bcd441f2b7f6243589b33da","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-19 03:03:11.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7372 : FAILURE in 1h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"705a37d736fe210cdc0c4ab79ed83b8853d3f050","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-19 04:01:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/3/check/dsvm-tempest-neutron-network/4616b77 : SUCCESS in 2h 01m 29s","accounts_in_message":[],"_revision_number":3},{"id":"acf099ace1dec8fa32bdbf22bbe28040764a3dd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 04:18:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/3/check/openstack-tox-pep8/c703453/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/3/check/openstack-tox-py27/5a58d0f/ : SUCCESS in 11m 17s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/3/check/openstack-tox-py35/877e07f/ : SUCCESS in 11m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/3/check/build-openstack-sphinx-docs/b430c9e/html/ : SUCCESS in 5m 41s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/3/check/legacy-tempest-dsvm-py35/37a678f/ : SUCCESS in 1h 20m 15s\n- legacy-gate-placement-api-ref http://logs.openstack.org/15/511715/3/check/legacy-gate-placement-api-ref/a993ca2/html/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ef38c83874a67e1a9397aa2ef4918cbf83dbe672","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 04:18:56.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"547e63146b71adf2291ede424fdf66f6cad02321","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 04:49:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/511715/3/ : SUCCESS in 1h 58m 09s","accounts_in_message":[],"_revision_number":3},{"id":"2756b5e6284726bc2573eeb7a86269828c908e66","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-24 09:00:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"206ba82480225730ae69c49e942837cda4912141","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-24 09:55:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/4/check/tempest-dsvm-full-xenial/8c88cbb/ : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":4},{"id":"dc646941fde6942f778d7eef1d1eb68fff3abe49","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-24 09:55:33.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":4},{"id":"5706377d558f55fa024f800a6de80c1794aee7ac","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-24 10:10:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/4/check/check-dsvm-tempest-vz7-exe-minimal/ec7445d : SUCCESS in 1h 05m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"96bfeeccfadb2be3da0f45b64dfafe7d48e8e06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-24 10:19:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/4/check/openstack-tox-pep8/1a099b5/ : SUCCESS in 8m 44s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/4/check/openstack-tox-py27/12b43a7/ : SUCCESS in 11m 31s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/4/check/openstack-tox-py35/a5014bf/ : SUCCESS in 11m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/4/check/build-openstack-sphinx-docs/a932ce0/html/ : SUCCESS in 5m 50s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/4/check/legacy-tempest-dsvm-py35/8987d80/ : SUCCESS in 1h 09m 39s\n- build-placement-api-ref http://logs.openstack.org/15/511715/4/check/build-placement-api-ref/eaf2e36/html/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":4},{"id":"00aff065e7ca69bfa7d4b527f9332c14872d08ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-24 10:19:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":4},{"id":"f9e9caa20f60fd7ede7c89724584c97de2a0324d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-24 10:58:41.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7526 : SUCCESS in 1h 57m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"999acfc5b8a9c25d2465f959ea6f601caa38d6a0","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-24 11:21:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/4/check/dsvm-tempest-neutron-network/ff0b26b : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"ffcfbe169ae777cc93095a9d432611197380c1bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-10-24 11:40:18.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-511715-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"58b30f9d19bb48b50ec92b4d2bf8df63a64bd8b9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-24 11:48:28.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/511715/4/ : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":4},{"id":"307027d88eb7c5902c92fac73de688c354822d30","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-24 12:25:38.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/4/check/nova-out-of-tree-pvm/d691ae7 : FAILURE in 3h 24m 13s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/4/check/nova-in-tree-pvm/2be6d95 : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":4},{"id":"365b2c292f422a54736ed899182a2fe49a4ba9f2","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-10-30 15:12:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47d34dde8a92d84203211aa3c6fab96b6a3f2faa","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-31 08:12:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d8b3ce55d257d284f6f2b794bcec2fdeab41c0d1","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-31 08:13:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0f7acffcd66a0379ef39bcbc52ae0f4e6939423a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-31 09:13:04.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8015 : FAILURE in 59m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e3024bd7429ba4ce80b28936b966dde3280c3b53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-31 09:17:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/5/check/tempest-dsvm-full-xenial/bddee86/ : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":5},{"id":"047788ee365b41d630f786fb814b9a380fa5a1c7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-31 09:18:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"ba560f5fb8df6c142a5aa39a81a3b2855f70fd52","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-31 09:22:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/5/check/nova-out-of-tree-pvm/de29e72 : SUCCESS in 1h 08m 28s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/5/check/nova-in-tree-pvm/536607d : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":5},{"id":"490f4e1788ae066ab1096f9a300f6835e5fe0595","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-31 09:24:49.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/5/check/check-dsvm-tempest-vz7-exe-minimal/e4bd391 : UNSTABLE in 1h 02m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"f8caf571855480f707e8525cfe8c2134adb71db8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-31 09:25:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/5/check/openstack-tox-pep8/37b430c/ : SUCCESS in 8m 11s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/5/check/openstack-tox-py27/97c60a7/ : SUCCESS in 10m 20s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/5/check/openstack-tox-py35/61b44a7/ : SUCCESS in 11m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/5/check/build-openstack-sphinx-docs/ce78a26/html/ : SUCCESS in 5m 05s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/5/check/legacy-tempest-dsvm-py35/0a6e607/ : SUCCESS in 1h 10m 36s\n- build-placement-api-ref http://logs.openstack.org/15/511715/5/check/build-placement-api-ref/ebc14dd/html/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":5},{"id":"03ee23702f55c28aaa29b10f8359a87481d60712","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-31 09:26:20.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"c0be3a15a81d8504832a8ca092e4b605599904c5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-31 10:12:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/5/check/dsvm-tempest-neutron-network/35d3f35 : SUCCESS in 1h 58m 12s","accounts_in_message":[],"_revision_number":5},{"id":"75f06545f7af7fb79404c8e99fa13b72c6382db9","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-10-31 10:28:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/15/511715/5/check/dsvm-tempest-xen/6d2e85c : SUCCESS in 2h 11m 03s","accounts_in_message":[],"_revision_number":5},{"id":"9f61dc8e6922c790de623ed8ca255bb324bdacf3","author":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"date":"2017-10-31 10:44:18.000000000","message":"Patch Set 5:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":5},{"id":"0ff1b314911f41327920a1ffb91b16ba83a38f4d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-31 11:23:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/15/511715/5/check-nova/Nova-ML2-Sriov/465b107 : SUCCESS in 1h 35m 03s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/15/511715/5/check-nova/Nova-MACVTAP-ML2-Sriov/f8eed51 : SUCCESS in 1h 09m 20s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"a17d85c09b0bcf9bd76eda5880d69957786f0a7f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-31 11:26:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/511715/5/ : SUCCESS in 3h 12m 50s","accounts_in_message":[],"_revision_number":5},{"id":"9d6da5b536ab5d6f19dba9afb6d660c9b9becdf9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-10-31 11:34:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"aa209bb6163f14ee908dc6d6e68a711898701fd5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-31 12:16:52.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8032 : SUCCESS in 1h 32m 17s","accounts_in_message":[],"_revision_number":5},{"id":"47a761e647e546acf3093395dc0be9fc3b572f37","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-11-27 17:44:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nThis is still a bit off, suggestion within.","accounts_in_message":[],"_revision_number":5},{"id":"fd0fbddfe3c37169a5de5443b56fc9640e6b9c0b","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-11-29 08:11:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a1d00226e3d2434d15f26929d64c626c9fc1700d","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-11-29 08:11:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4dd43c3587f74e8c2321098e92388ef8d672bfbd","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-29 09:07:45.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/6/check/nova-out-of-tree-pvm/0b8b10a : SUCCESS in 56m 08s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/6/check/nova-in-tree-pvm/79c7d7a : FAILURE in 38m 53s","accounts_in_message":[],"_revision_number":6},{"id":"3351e5cb2ce0b965e25390ee65bddd6459f6a867","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-11-29 09:14:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/6/check/tempest-dsvm-full-xenial/15ddabf/ : SUCCESS in 1h 02m 53s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/6/check/grenade-dsvm-xenial/f69da55/ : SUCCESS in 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"059ae4ca0c047e166dc39de3dc404d225e21fea7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-11-29 09:27:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/6/check/check-dsvm-tempest-vz7-exe-minimal/808b020 : SUCCESS in 57m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"f8f9eb7839b10c17e573b8680dc1d77785131880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 09:29:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/6/check/openstack-tox-pep8/91f7ef1/ : SUCCESS in 8m 16s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/6/check/openstack-tox-py27/c3ed3ec/ : SUCCESS in 16m 47s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/6/check/openstack-tox-py35/d7166a6/ : SUCCESS in 11m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/6/check/build-openstack-sphinx-docs/8a8373c/html/ : SUCCESS in 5m 25s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/6/check/legacy-tempest-dsvm-py35/8a89786/ : SUCCESS in 1h 16m 00s\n- build-placement-api-ref http://logs.openstack.org/15/511715/6/check/build-placement-api-ref/9282d41/html/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":6},{"id":"8236f4662d586de57861625b35225e65ab2c4e0f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-11-29 10:18:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/15/511715/6/check/dsvm-tempest-neutron-network/be7f3cf : SUCCESS in 2h 06m 57s","accounts_in_message":[],"_revision_number":6},{"id":"6b993af177eae9d2adb517d38c3bd005e6b7a21b","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-11-29 10:37:59.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/15/511715/6/check/dsvm-tempest-xen/1a022cb : FAILURE in 2h 26m 06s","accounts_in_message":[],"_revision_number":6},{"id":"fb9a83adb61e358ebcc717b8a344ec9a9bea8cdd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-11-29 12:49:30.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-15-511715-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"bcaef751497050f684bebb3cd5132fd7003c6449","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-11-29 17:57:53.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/15/511715/6/check-nova/Nova-ML2-Sriov/810df69 : FAILURE in 1h 20m 40s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/15/511715/6/check-nova/Nova-MACVTAP-ML2-Sriov/97f478b : FAILURE in 1h 17m 58s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"32a1cc29b09b7d6e79c86396b85763e5eb4e3370","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-11-30 01:36:06.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/511715/6/ : FAILURE in 6h 30m 35s","accounts_in_message":[],"_revision_number":6},{"id":"56525c8e7e119824b8dc37be73b5d326ef51b381","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-11-30 02:43:47.000000000","message":"Patch Set 6: Code-Review+1\n\nIt still looks good to me.","accounts_in_message":[],"_revision_number":6},{"id":"c8bcaaf0c1eda60e9501d30d8fd4d1e8c56f35f0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 20:35:37.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab988a2b5a47cd3563364852c2eeea3741f2cc7f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 20:41:19.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a8245806a51e47759ebd2c8bc8536c208c89a5d2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-30 20:41:26.000000000","message":"Patch Set 7: Code-Review+2\n\nYup:\n\nhttps://github.com/openstack/nova/blob/master/nova/api/openstack/placement/handlers/allocation.py#L251","accounts_in_message":[],"_revision_number":7},{"id":"4034ffc79644e4208e3979abf659b8260cbb53a0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-30 21:03:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c9921961d8886e0de7b8602040ad42f102706365","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-11-30 21:35:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/7/check/tempest-dsvm-full-xenial/b3b3d18/ : SUCCESS in 53m 13s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/15/511715/7/check/grenade-dsvm-xenial/64cde89/ : SUCCESS in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3a540b65d9e259e0401c86d6529d078bbc752069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 22:01:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/7/check/openstack-tox-pep8/2562e22/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/7/check/openstack-tox-py27/2b2a19d/ : SUCCESS in 11m 39s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/7/check/openstack-tox-py35/4909743/ : SUCCESS in 12m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/7/check/build-openstack-sphinx-docs/f5d4292/html/ : SUCCESS in 5m 07s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/7/check/legacy-tempest-dsvm-py35/29cd3ec/ : SUCCESS in 1h 13m 59s\n- build-placement-api-ref http://logs.openstack.org/15/511715/7/check/build-placement-api-ref/e985395/html/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":7},{"id":"edc5fea8047bf0b99f991dbd29bfc07fb27a33c8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-30 22:06:21.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/15/511715/7/check/nova-out-of-tree-pvm/1dc7928 : SUCCESS in 1h 02m 35s\n- nova-in-tree-pvm http://184.172.12.213/15/511715/7/check/nova-in-tree-pvm/5c14b51 : FAILURE in 34m 16s","accounts_in_message":[],"_revision_number":7},{"id":"2c495f2b084e840ebbb08669565139244fda4d6e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-11-30 22:25:14.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/15/511715/7/check/check-dsvm-tempest-vz7-exe-minimal/0190088 : SUCCESS in 51m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"a66d3985983285b68a06b27fcc69dbb0d159fd2f","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-11-30 22:48:54.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/15/511715/7/check/dsvm-tempest-xen/38f84ca : FAILURE in 2h 06m 45s","accounts_in_message":[],"_revision_number":7},{"id":"144bf4281eb7d4536fb19457508fb574cf21c106","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-01 00:59:46.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/15/511715/7/check-nova/Nova-ML2-Sriov/047e9fa : FAILURE in 51m 11s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/15/511715/7/check-nova/Nova-MACVTAP-ML2-Sriov/7ac37fe : FAILURE in 1h 06m 25s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"59b1dc4db2a9099ec8bddcf668b533190f8e0394","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-01 06:00:06.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"8b34a9355dfd35fc35241fceb430cfb6996db32a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-01 09:51:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/511715/7/ : SUCCESS in 3h 26m 34s","accounts_in_message":[],"_revision_number":7},{"id":"02b1914b1ab158664a56e4815acc5801edb15cde","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-12-05 11:37:33.000000000","message":"Patch Set 7: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"56de67916a2a8f94b41732b85fb96b975a333be2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 11:40:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"30bfb94d4ebaaf8f119734734a87e256b5422374","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 13:32:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"8ae9f141d9e61528fb39cee294424e57d7dc91cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 13:32:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/15/511715/7/gate/openstack-tox-pep8/18b4c0f/ : SUCCESS in 7m 43s\n- openstack-tox-py27 http://logs.openstack.org/15/511715/7/gate/openstack-tox-py27/a6b0175/ : SUCCESS in 11m 06s\n- openstack-tox-py35 http://logs.openstack.org/15/511715/7/gate/openstack-tox-py35/ce9bbd0/ : SUCCESS in 12m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/15/511715/7/gate/build-openstack-sphinx-docs/6a41d97/html/ : SUCCESS in 5m 37s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/15/511715/7/gate/legacy-tempest-dsvm-py35/75f2fa5/ : SUCCESS in 1h 24m 00s\n- build-placement-api-ref http://logs.openstack.org/15/511715/7/gate/build-placement-api-ref/6a6d619/html/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1753195c75ebc7380bccc8b39c8a16940fa66658","revisions":{"245ae9bb733c97d38ec652f66edee46265478ec0":{"kind":"REWORK","_number":1,"created":"2017-10-13 05:48:05.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/1"}}},"commit":{"parents":[{"commit":"9b62f5d01926c877961f2fc26d5d95ad8f892689","subject":"Merge \"Make TestRPC inherit from the base nova TestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b62f5d01926c877961f2fc26d5d95ad8f892689"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-09-26 20:23:13.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-09-26 20:23:13.000000000","tz":480},"subject":"Enhance doc for placement allocation list","message":"Enhance doc for placement allocation list\n\n1) add a note about when the non-existing consumer id input result\n2) only keep one consumer id result in the sample file as the\nallocation only target to one consumer\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/245ae9bb733c97d38ec652f66edee46265478ec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/245ae9bb733c97d38ec652f66edee46265478ec0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b62f5d01926c877961f2fc26d5d95ad8f892689","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"259f6738c722243548d04e162342f4469c997e72":{"kind":"REWORK","_number":2,"created":"2017-10-13 07:53:04.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/2"}}},"commit":{"parents":[{"commit":"9b62f5d01926c877961f2fc26d5d95ad8f892689","subject":"Merge \"Make TestRPC inherit from the base nova TestCase\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b62f5d01926c877961f2fc26d5d95ad8f892689"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-09-26 20:23:13.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-09-26 22:28:42.000000000","tz":480},"subject":"Enhance doc for placement allocation list","message":"Enhance doc for placement allocation list\n\n1) add a note about when the non-existing consumer id input result\n2) only keep one consumer id result in the sample file as the\nallocation only target to one consumer\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/259f6738c722243548d04e162342f4469c997e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/259f6738c722243548d04e162342f4469c997e72"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9b62f5d01926c877961f2fc26d5d95ad8f892689","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5d5e0deb921731e4211a3b66fde4de16effc7e99":{"kind":"REWORK","_number":3,"created":"2017-10-19 01:59:38.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/3"}}},"commit":{"parents":[{"commit":"e11a8687aef527eee9f7c733db6cccd5b902afbb","subject":"Merge \"fix cleaning up evacuated instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e11a8687aef527eee9f7c733db6cccd5b902afbb"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"subject":"Enhance doc for placement allocation list","message":"Enhance doc for placement allocation list\n\nadd a note about when the non-existing consumer id input result\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d5e0deb921731e4211a3b66fde4de16effc7e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d5e0deb921731e4211a3b66fde4de16effc7e99"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e11a8687aef527eee9f7c733db6cccd5b902afbb","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6dcd836e27875b7384ca8cf45beeb12efd57c5d9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2017-10-24 09:00:56.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/4"}}},"commit":{"parents":[{"commit":"02dce6fc43ff7a2f444d01369315ae0f97cf39a1","subject":"Merge \"conf: Remove deprecated \u0027null_kernel\u0027 opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02dce6fc43ff7a2f444d01369315ae0f97cf39a1"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-20 00:44:54.000000000","tz":480},"subject":"[placement]Enhance doc for placement allocation list","message":"[placement]Enhance doc for placement allocation list\n\nadd a note about when the non-existing consumer id input result\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6dcd836e27875b7384ca8cf45beeb12efd57c5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6dcd836e27875b7384ca8cf45beeb12efd57c5d9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"02dce6fc43ff7a2f444d01369315ae0f97cf39a1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8d8e5f5cccbf2716208faef6bc0235df90fe3fac":{"kind":"REWORK","_number":5,"created":"2017-10-31 08:13:11.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/5"}}},"commit":{"parents":[{"commit":"02dce6fc43ff7a2f444d01369315ae0f97cf39a1","subject":"Merge \"conf: Remove deprecated \u0027null_kernel\u0027 opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02dce6fc43ff7a2f444d01369315ae0f97cf39a1"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-31 08:12:31.000000000","tz":480},"subject":"[placement]Enhance doc for placement allocation list","message":"[placement]Enhance doc for placement allocation list\n\nadd a note about when the non-existing consumer id input result\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d8e5f5cccbf2716208faef6bc0235df90fe3fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d8e5f5cccbf2716208faef6bc0235df90fe3fac"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"02dce6fc43ff7a2f444d01369315ae0f97cf39a1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"60384f4907ee0df8d50c8f7e600e4fd5748060de":{"kind":"REWORK","_number":6,"created":"2017-11-29 08:11:11.000000000","uploader":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"ref":"refs/changes/15/511715/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/6"}}},"commit":{"parents":[{"commit":"7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd","subject":"Merge \"PowerVM support matrix update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"committer":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-11-28 17:46:13.000000000","tz":480},"subject":"[placement]Enhance doc for placement allocation list","message":"[placement]Enhance doc for placement allocation list\n\nadd a note about when the non-existing consumer id input result\n\nPartial implement blueprint placement-doc-enhancement-queens\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60384f4907ee0df8d50c8f7e600e4fd5748060de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60384f4907ee0df8d50c8f7e600e4fd5748060de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1753195c75ebc7380bccc8b39c8a16940fa66658":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2017-11-30 20:41:19.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/15/511715/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/511715/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/511715/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/511715/7"}}},"commit":{"parents":[{"commit":"7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd","subject":"Merge \"PowerVM support matrix update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd"}]}],"author":{"name":"jichenjc","email":"jichenjc@cn.ibm.com","date":"2017-10-18 14:34:19.000000000","tz":480},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2017-11-30 20:41:08.000000000","tz":0},"subject":"[placement]Enhance doc for placement allocation list","message":"[placement]Enhance doc for placement allocation list\n\nadd a note about when the non-existing consumer id input result\n\nChange-Id: I2a19d8fe8dda5754112361be5ceac8a0dc747ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1753195c75ebc7380bccc8b39c8a16940fa66658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1753195c75ebc7380bccc8b39c8a16940fa66658"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7e2d8cb89b68d136c7bc3dd2eae18b5269fcefbd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
