)]}'
{"id":"openstack%2Fnova~512329","triplet_id":"openstack%2Fnova~master~Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e","project":"openstack/nova","branch":"master","topic":"blow-up-if-service-user-requested-but-no-auth","hashtags":[],"change_id":"Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e","subject":"Service user token requested with no auth","status":"ABANDONED","created":"2017-10-16 15:27:48.000000000","updated":"2017-11-21 17:31:28.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e3a83c90913d862d5df4bc6f33fc421fa959111","_number":512329,"virtual_id_number":512329,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"all":[{"date":"2017-10-19 22:19:35.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"date":"2017-10-19 21:44:55.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-19 17:44:26.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2017-10-19 19:55:43.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2017-10-20 07:33:24.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-19 17:17:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2017-10-19 17:57:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":1,"date":"2017-10-19 22:20:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"date":"2017-10-20 05:57:15.000000000","_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"date":"2017-10-19 18:15:26.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2017-10-19 23:08:14.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2017-10-19 20:43:08.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":1,"date":"2017-10-20 09:44:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2017-11-21 17:06:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2017-11-21 16:38:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-17 11:58:02.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2017-10-18 21:35:03.000000000","updated_by":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"reviewer":{"_account_id":26463,"name":"Niraj Singh","email":"niraj.singh@nttdata.com","username":"nirajsingh"},"state":"REVIEWER"},{"updated":"2017-10-19 17:17:02.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 17:44:26.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 17:57:36.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-10-19 18:15:26.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 19:55:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 21:44:55.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 22:19:35.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 22:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-19 23:08:14.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-10-20 05:57:15.000000000","updated_by":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"reviewer":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"state":"REVIEWER"},{"updated":"2017-10-20 07:33:24.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-20 09:44:30.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2017-11-21 16:38:40.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2017-11-21 17:06:41.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"}],"messages":[{"id":"30da41a603b8b7f79bedc25c1c39e7a33d8d2a8a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-16 15:27:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c06c7aefb2aa534ae8e7d130e8698897e433d548","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-16 16:23:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/1/check/tempest-dsvm-full-xenial/b5c40ec/ : SUCCESS in 55m 17s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/1/check/grenade-dsvm-xenial/63b57ce/ : SUCCESS in 48m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aac92bb8f3a937eede10d7a7780388bbdc48f35e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-16 16:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/29/512329/1/check/check-dsvm-tempest-vz7-exe-minimal/88770fd : SUCCESS in 1h 06m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"c289d79e7b3a3d3880738c0af6643b2345328c77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 16:47:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/29/512329/1/check/openstack-tox-pep8/4e21399/ : SUCCESS in 7m 51s\n- openstack-tox-py27 http://logs.openstack.org/29/512329/1/check/openstack-tox-py27/28988a7/ : FAILURE in 11m 48s\n- openstack-tox-py35 http://logs.openstack.org/29/512329/1/check/openstack-tox-py35/d6f5ee4/ : FAILURE in 12m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/512329/1/check/build-openstack-sphinx-docs/d6f36bb/html/ : SUCCESS in 5m 06s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-py35/2fb325e/ : SUCCESS in 1h 11m 10s\n- legacy-nova-tox-functional http://logs.openstack.org/29/512329/1/check/legacy-nova-tox-functional/2316857/ : SUCCESS in 23m 23s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/29/512329/1/check/legacy-nova-tox-functional-py35/7e2e314/ : SUCCESS in 17m 34s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/29/512329/1/check/legacy-grenade-dsvm-neutron-multinode/9b5b78f/ : SUCCESS in 1h 12m 28s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/29/512329/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/8cfd49c/ : FAILURE in 54m 26s (non-voting)\n- legacy-tempest-dsvm-cells legacy-tempest-dsvm-cells : ERROR Project openstack/cinder does not have the default branch master\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/44f843d/ : SUCCESS in 1h 08m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-neutron-multinode-full/9c1d66c/ : FAILURE in 1h 17m 35s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-neutron-nova-next-full/9dfb1e3/ : SUCCESS in 1h 14m 51s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-multinode-live-migration/756e7db/ : SUCCESS in 43m 38s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/29/512329/1/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3062bb3/ : SUCCESS in 42m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-full legacy-tempest-dsvm-neutron-full : ERROR Project openstack/cinder does not have the default branch master\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/29/512329/1/check/legacy-grenade-dsvm-neutron/2630a86/ : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f15c66d38749c41874dfd6f3251a27a542ff9ab3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-16 17:04:28.000000000","message":"Patch Set 1: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/512329/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f446de2 : SUCCESS in 1h 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ee6fd446974c6f8b8c6423e5f64e3aa513268da","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-16 17:07:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d0a168194e2751d4dc18d717a92d48e536441045","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-16 18:04:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/2/check/tempest-dsvm-full-xenial/c506384/ : SUCCESS in 55m 51s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/2/check/grenade-dsvm-xenial/7a3f7a4/ : SUCCESS in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"193c71a4525aeb01f0bda5ca24a8adddc26aed08","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-16 18:30:37.000000000","message":"Patch Set 2: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7201 : SUCCESS in 1h 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1062b765c5fb771983bb8e6fe2e3a5933af82347","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-16 18:41:49.000000000","message":"Patch Set 2: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/512329/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2fdcbed : SUCCESS in 1h 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b227f06a2aa16b883bfc7ad4390f3c1a4acaab82","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-16 18:43:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/29/512329/2/check/check-dsvm-tempest-vz7-exe-minimal/62431c3 : SUCCESS in 1h 08m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"36af2be3a0b40e3033fc756c0df22cec261584dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 19:05:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/29/512329/2/check/openstack-tox-pep8/fa14fd2/ : SUCCESS in 9m 41s\n- openstack-tox-py27 http://logs.openstack.org/29/512329/2/check/openstack-tox-py27/c1032c6/ : SUCCESS in 13m 04s\n- openstack-tox-py35 http://logs.openstack.org/29/512329/2/check/openstack-tox-py35/19f0d84/ : SUCCESS in 11m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/512329/2/check/build-openstack-sphinx-docs/2703bf1/html/ : SUCCESS in 6m 49s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-py35/417834e/ : SUCCESS in 1h 17m 41s\n- legacy-nova-tox-functional http://logs.openstack.org/29/512329/2/check/legacy-nova-tox-functional/da7ee2d/ : SUCCESS in 21m 44s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/29/512329/2/check/legacy-nova-tox-functional-py35/e74f1c5/ : SUCCESS in 19m 19s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/29/512329/2/check/legacy-grenade-dsvm-neutron-multinode/6701a3a/ : SUCCESS in 1h 00m 57s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/29/512329/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/c448b08/ : FAILURE in 1h 04m 32s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-cells/5962e65/ : SUCCESS in 34m 18s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6b64b35/ : SUCCESS in 1h 07m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-neutron-multinode-full/665b7cd/ : FAILURE in 1h 21m 55s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-neutron-nova-next-full/badafd2/ : SUCCESS in 1h 10m 44s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-multinode-live-migration/ade11cc/ : SUCCESS in 42m 47s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d4312dd/ : SUCCESS in 48m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/512329/2/check/legacy-tempest-dsvm-neutron-full/da8093d/ : SUCCESS in 1h 41m 56s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/29/512329/2/check/legacy-grenade-dsvm-neutron/e9e8b52/ : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":2},{"id":"cd63641b80961314c325c9de362117b12086de61","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-16 19:14:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/29/512329/2/check/dsvm-tempest-neutron-network/635d2b5 : SUCCESS in 2h 05m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2efcd9c82b30649b589818e084e277a33b72ef25","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-16 19:41:57.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/29/512329/2/check/nova-out-of-tree-pvm/5fc8ecb : FAILURE in 1h 38m 37s\n- nova-in-tree-pvm http://184.172.12.213/29/512329/2/check/nova-in-tree-pvm/e1ce57e : FAILURE in 44m 31s","accounts_in_message":[],"_revision_number":2},{"id":"abaa3b2268a90183212f8eb08641215d1446e5fd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-16 21:53:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/512329/2/ : SUCCESS in 2h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8584fed986d3dfb174242a2130aeeeeae87bc3e3","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2017-10-17 11:58:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9f7e61fdd8689510c8674c26173506e5ffe66ad2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-18 21:30:58.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b4e6226b6fb03cd11a99067ae6de4ab381a4e95f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-18 21:34:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"09899436558fa96c833e7dcaf2e483c5cc244f9a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-18 22:33:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/29/512329/3/check/check-dsvm-tempest-vz7-exe-minimal/01c8013 : SUCCESS in 1h 01m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"3915e67f639c9e2c6463a83879a91f944880f27e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-18 22:33:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/3/check/tempest-dsvm-full-xenial/9ed0d14/ : SUCCESS in 1h 01m 54s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/3/check/grenade-dsvm-xenial/0e2d3e3/ : FAILURE in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a16b428a307ee8931b57082889d4f8d8076b326","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-18 23:19:20.000000000","message":"Patch Set 3: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/512329/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/68dc8b5 : SUCCESS in 1h 47m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1bc5a84fb0bb8435826d8b60c93442260418b8fc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-18 23:25:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/29/512329/3/check/dsvm-tempest-neutron-network/8e083ad : SUCCESS in 1h 53m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ca5f5370e452e763b5429cb845fab98836668ff7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-18 23:25:23.000000000","message":"Patch Set 3: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7359 : FAILURE in 1h 54m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"972b148f2350e62daa282288f4b4327b92e3f6ef","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 00:06:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/29/512329/3/check/nova-out-of-tree-pvm/1d02d6b : SUCCESS in 2h 33m 37s\n- nova-in-tree-pvm http://184.172.12.213/29/512329/3/check/nova-in-tree-pvm/4c639ae : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2936455d223bd3a1c990fde1dd050e9aad017d09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 02:45:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/29/512329/3/check/openstack-tox-pep8/dfa3693/ : SUCCESS in 8m 43s\n- openstack-tox-py27 http://logs.openstack.org/29/512329/3/check/openstack-tox-py27/524f40f/ : FAILURE in 12m 20s\n- openstack-tox-py35 http://logs.openstack.org/29/512329/3/check/openstack-tox-py35/f746f63/ : FAILURE in 11m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/512329/3/check/build-openstack-sphinx-docs/caacc7c/html/ : SUCCESS in 4m 58s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-py35/7e9cfe0/ : SUCCESS in 1h 22m 04s\n- legacy-nova-tox-functional http://logs.openstack.org/29/512329/3/check/legacy-nova-tox-functional/64ddf10/ : SUCCESS in 18m 19s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/29/512329/3/check/legacy-nova-tox-functional-py35/d6283be/ : SUCCESS in 21m 08s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/29/512329/3/check/legacy-grenade-dsvm-neutron-multinode/673c299/ : SUCCESS in 1h 05m 01s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/29/512329/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/de6c9f8/ : FAILURE in 1h 09m 11s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-cells/25de88b/ : SUCCESS in 44m 52s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/e3aee1d/ : SUCCESS in 1h 18m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-neutron-multinode-full/f82071b/ : SUCCESS in 1h 39m 23s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-neutron-nova-next-full/a035536/ : SUCCESS in 1h 29m 11s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-multinode-live-migration/ccf3322/ : SUCCESS in 48m 36s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0363a52/ : SUCCESS in 45m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/512329/3/check/legacy-tempest-dsvm-neutron-full/b61e14f/ : SUCCESS in 1h 25m 05s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/29/512329/3/check/legacy-grenade-dsvm-neutron/6fae72b/ : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":3},{"id":"49a8b62b2384266a197fa8ae4119c930fd780f3e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 04:03:31.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/512329/3/ : FAILURE in 5h 36m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5e4019db3271d61bb55af5d016402bc24ef90931","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-19 16:12:10.000000000","message":"Patch Set 3:\n\nrecheck\n\nThat\u0027s a weird one...","accounts_in_message":[],"_revision_number":3},{"id":"1d2ed2120f17c7c89017bc10c02350aff6ee6dae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-19 16:14:19.000000000","message":"Patch Set 3: Workflow-1\n\nOh.  Globals.  Yeah, that should be fixed.","accounts_in_message":[],"_revision_number":3},{"id":"189fe444620f840aa4d5b6de439a77fb84be4f4c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-19 16:18:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"62c0cfe366de622ec53954fa39df68b58fa0be94","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-19 16:21:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e00eaf7e5ff9d2ddf55100fe0334ce361120924c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-19 17:17:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/5/check/tempest-dsvm-full-xenial/b2a5172/ : SUCCESS in 54m 48s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/5/check/grenade-dsvm-xenial/372590a/ : FAILURE in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf2d6c327a2928e1a17ec2d56e09e72e266ab042","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 17:17:32.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"a8bbdb98b12bafcd0decba926489c9de448d7c01","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 17:44:26.000000000","message":"Patch Set 5: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/512329/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/35323cb : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":5},{"id":"d7dceabaef5963088c2dc29c4f95eff539ab56aa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 17:44:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"baf90afd1e6075718154fab39afe70d72a22a435","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 17:54:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/29/512329/5/check/check-dsvm-tempest-vz7-exe-minimal/6df8c89 : SUCCESS in 1h 26m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"4a53300e398af308a702d3f2dfdf555a9da201a2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-19 17:57:36.000000000","message":"Patch Set 5: Verified+1\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7406 : FAILURE in 1h 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f53c8fbd04be0cf6bd6f01ff2c0d2d9df13cdc9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 18:00:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/29/512329/5/check/nova-out-of-tree-pvm/3a2d5db : SUCCESS in 1h 38m 00s\n- nova-in-tree-pvm http://184.172.12.213/29/512329/5/check/nova-in-tree-pvm/000270c : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":5},{"id":"2f6f38bc50ac4695a38bcbcbd41f1d6583f55f98","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-10-19 18:15:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/29/512329/5/check/dsvm-tempest-neutron-network/2043148 : SUCCESS in 1h 53m 16s","accounts_in_message":[],"_revision_number":5},{"id":"331937ed99752b146511db869dd91b9c627411d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 19:51:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/29/512329/5/check/openstack-tox-pep8/3a1c44c/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/29/512329/5/check/openstack-tox-py27/54825a7/ : SUCCESS in 13m 02s\n- openstack-tox-py35 http://logs.openstack.org/29/512329/5/check/openstack-tox-py35/4327ae3/ : SUCCESS in 10m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/512329/5/check/build-openstack-sphinx-docs/6fd6c5f/html/ : SUCCESS in 6m 16s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-py35/0751d0c/ : SUCCESS in 1h 33m 44s\n- legacy-nova-tox-functional http://logs.openstack.org/29/512329/5/check/legacy-nova-tox-functional/e1496c8/ : SUCCESS in 25m 28s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/29/512329/5/check/legacy-nova-tox-functional-py35/239dccc/ : SUCCESS in 21m 19s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron-multinode/55affcc/ : SUCCESS in 1h 04m 43s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/9e5e9b8/ : FAILURE in 1h 07m 39s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-cells/70e9fa4/ : SUCCESS in 45m 39s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/4aa5062/ : SUCCESS in 1h 12m 37s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-multinode-full/919ac2a/ : SUCCESS in 1h 32m 30s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-nova-next-full/6e4f0b9/ : SUCCESS in 1h 25m 15s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-multinode-live-migration/8b1ae68/ : SUCCESS in 44m 22s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/357eb30/ : SUCCESS in 51m 24s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-full/8d590ff/ : FAILURE in 1h 13m 17s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron/bfd85ef/ : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":5},{"id":"76d541304ec643dce66dcb23aced89d6be4e0909","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-10-19 19:55:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/29/512329/5/check-nova/Nova-ML2-Sriov/1bfd60a : SUCCESS in 57m 33s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/29/512329/5/check-nova/Nova-MACVTAP-ML2-Sriov/ffcfdf9 : SUCCESS in 44m 26s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"a7ce5f29709696b091235ead4920a30a49862cfe","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 20:05:36.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/512329/5/ : FAILURE in 3h 43m 33s","accounts_in_message":[],"_revision_number":5},{"id":"97007601f7067202f02426c6dee15a7524135388","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-10-19 20:43:08.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e3d281d24fa6c1b200fb2f5bc3aed5b051ce7b33","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-10-19 21:41:16.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/5/check/tempest-dsvm-full-xenial/75f5ead/ : SUCCESS in 57m 23s\n- grenade-dsvm-xenial http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/29/512329/5/check/grenade-dsvm-xenial/e6d7dcc/ : FAILURE in 44m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9f9864342c2a229502289d9b28fcd09b7ee2c08e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 21:41:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"1c81a1a673a26aa8674f8e04866e59aafa84609b","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 21:44:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/29/512329/5/check/check-dsvm-tempest-vz7-exe-minimal/27779a3 : SUCCESS in 1h 01m 09s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"344518a314279bc767281d5cbff6fd1a1a1251bc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-19 22:11:53.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7414 : FAILURE in 1h 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0f0ed901443e0a6baa375b7e5d0587ccffb7886","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 22:12:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"107135bb66bd1da2bb8dd1cab5dd8bb38aaed122","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-10-19 22:19:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/29/512329/5/check/nova-out-of-tree-pvm/e0f355f : SUCCESS in 1h 31m 41s\n- nova-in-tree-pvm http://184.172.12.213/29/512329/5/check/nova-in-tree-pvm/99d34b5 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":5},{"id":"1d59325f1a2b0b747de09ca682b16806058ab554","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-19 22:20:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/29/512329/5/check/openstack-tox-pep8/ec5666c/ : SUCCESS in 8m 51s\n- openstack-tox-py27 http://logs.openstack.org/29/512329/5/check/openstack-tox-py27/fbe9c67/ : SUCCESS in 11m 51s\n- openstack-tox-py35 http://logs.openstack.org/29/512329/5/check/openstack-tox-py35/b9ac260/ : SUCCESS in 11m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/29/512329/5/check/build-openstack-sphinx-docs/be08a0c/html/ : SUCCESS in 6m 12s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-py35/89060fd/ : SUCCESS in 1h 16m 34s\n- legacy-nova-tox-functional http://logs.openstack.org/29/512329/5/check/legacy-nova-tox-functional/dd78277/ : SUCCESS in 20m 42s\n- legacy-nova-tox-functional-py35 http://logs.openstack.org/29/512329/5/check/legacy-nova-tox-functional-py35/0d80310/ : SUCCESS in 16m 58s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron-multinode/743cf66/ : SUCCESS in 1h 01m 09s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/e32bf95/ : FAILURE in 1h 17m 14s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-cells/e4d8a6e/ : SUCCESS in 43m 39s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d75ed56/ : SUCCESS in 1h 04m 13s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-multinode-full/41b807b/ : SUCCESS in 1h 34m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-nova-next-full/52b28d2/ : SUCCESS in 1h 19m 11s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-multinode-live-migration/540be51/ : SUCCESS in 46m 25s\n- legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b93006d/ : SUCCESS in 47m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/512329/5/check/legacy-tempest-dsvm-neutron-full/ae751e8/ : SUCCESS in 1h 17m 48s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/29/512329/5/check/legacy-grenade-dsvm-neutron/c2d4284/ : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":5},{"id":"5ea395b36bff105bf8a9c2913d275d798c131ed8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-19 22:20:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"32dde62c43b05016e4cfa844a8d2f2e81bf13785","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-10-19 23:08:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/512329/5/ : SUCCESS in 2h 24m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ae55a06f91a3a4367657d9f08911de2eca60a7bd","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2017-10-20 05:57:15.000000000","message":"Patch Set 5:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":5},{"id":"24a8fbbc3e720993acbcf335df8ff24b785c3dee","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-10-20 07:32:59.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-7424 : FAILURE in 1h 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c29ab62c10625939551cb7015e26249763ead77","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-10-20 07:33:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":5},{"id":"77ea987f2d586fa04f94e5c10ba1e9fae4b20631","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2017-10-20 09:44:30.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"35105154c21b9688b90f8f41c69f9f5983850e68","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2017-11-21 16:38:40.000000000","message":"Patch Set 5: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":5},{"id":"19483227e62622a89f1d61ffd2714a921236ce44","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-21 17:06:41.000000000","message":"Patch Set 5: Code-Review-1\n\nI didn\u0027t know this patch existed, but I wrote my own which is pretty close:\n\nhttps://review.openstack.org/521947\n\nI prefer mine since it doesn\u0027t punish the end user because the operator screwed up. In this change, the user operation fails outright. In mine, you at least get a chance of succeeding and we log a warning for the operator in case it eventually fails to re-auth (and the user opens a support ticket).","accounts_in_message":[],"_revision_number":5},{"id":"6ea540e078ba180f4b81914e67cb8667a48046b4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-21 17:09:25.000000000","message":"Patch Set 5:\n\nThis also probably generates a bunch of places to raise a 500 error out of the REST API to the end user, for an operator issue.","accounts_in_message":[],"_revision_number":5},{"id":"775ba5eaf0c22af0a2be765fdc2a25eeb240524f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-11-21 17:14:31.000000000","message":"Patch Set 5:\n\nFWIW I also wouldn\u0027t backport this to stable branches because I wouldn\u0027t expect operators picking up stable fixes to be very happy if all of a sudden they start getting 500s out of the API for something that\u0027s been \"working\" until the point of getting the stable patch release. I get that we don\u0027t like to silently fail, but punching the user in the face because the operator screwed up, and we have a reasonable chance of not blowing up in 95% of cases (no token timeout) then we should just fallback to that. If users complain of token timeout issues to their cloud support team and they then investigate why the service_user stuff isn\u0027t working, they would see the warnings in the logs from my patch.","accounts_in_message":[],"_revision_number":5},{"id":"8b8807f5ca015ee1665121f8690774aeac808132","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-21 17:31:28.000000000","message":"Abandoned\n\nSuperseded by https://review.openstack.org/521947","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cdaee2e114e4db886ea6eb1964df50020289e0fa","revisions":{"c03317f6b57930521ecd7a2e3e82b40d2628d7cc":{"kind":"REWORK","_number":1,"created":"2017-10-16 15:27:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/29/512329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/29/512329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/29/512329/1"}}},"commit":{"parents":[{"commit":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","subject":"Merge \"Refactor init_instance:resume_guests_state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1148a2d67b70dd767f798e1dfe0b4c7634d2f90c"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"subject":"Service user token requested with no auth","message":"Service user token requested with no auth\n\nIf [service_user]send_service_user_token is requested, but no auth\ninformation is provided in the [service_user] section, the resulting\nerror is cryptic [1] and doesn\u0027t help the user understand what went\nwrong.\n\nWith this change, we raise an exception with a helpful message if the\n[service_user]send_service_user_token option is requested, but the ksa\nrequest to load the service auth returns None.\n\nNote that today this only happens if the conf section is missing the\nauth_type option.  But this is a pretty good first indicator that the\nadmin forgot to populate auth options in general.\n\n[1] http://paste.openstack.org/show/623721/\n\nChange-Id: Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c03317f6b57930521ecd7a2e3e82b40d2628d7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c03317f6b57930521ecd7a2e3e82b40d2628d7cc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"88b8e4a3ad1ce9030714d7d7f68a32b0aa0eb501":{"kind":"REWORK","_number":2,"created":"2017-10-16 17:07:30.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/29/512329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/29/512329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/29/512329/2"}}},"commit":{"parents":[{"commit":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","subject":"Merge \"Refactor init_instance:resume_guests_state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1148a2d67b70dd767f798e1dfe0b4c7634d2f90c"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 17:07:28.000000000","tz":-300},"subject":"Service user token requested with no auth","message":"Service user token requested with no auth\n\nIf [service_user]send_service_user_token is requested, but no auth\ninformation is provided in the [service_user] section, the resulting\nerror is cryptic [1] and doesn\u0027t help the user understand what went\nwrong.\n\nWith this change, we raise an exception with a helpful message if the\n[service_user]send_service_user_token option is requested, but the ksa\nrequest to load the service auth returns None.\n\nNote that today this only happens if the conf section is missing the\nauth_type option.  But this is a pretty good first indicator that the\nadmin forgot to populate auth options in general.\n\n[1] http://paste.openstack.org/show/623721/\n\nChange-Id: Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88b8e4a3ad1ce9030714d7d7f68a32b0aa0eb501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88b8e4a3ad1ce9030714d7d7f68a32b0aa0eb501"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8c90079a8890c7f294b654bb1d15fd3de9f2322f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2017-10-18 21:30:58.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/29/512329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/29/512329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/29/512329/3"}}},"commit":{"parents":[{"commit":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","subject":"Merge \"Refactor init_instance:resume_guests_state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1148a2d67b70dd767f798e1dfe0b4c7634d2f90c"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-18 21:30:51.000000000","tz":0},"subject":"Service user token requested with no auth","message":"Service user token requested with no auth\n\nIf [service_user]send_service_user_token is requested, but no auth\ninformation is provided in the [service_user] section, the resulting\nerror is cryptic [1] and doesn\u0027t help the user understand what went\nwrong.\n\nWith this change, we raise an exception with a helpful message if the\n[service_user]send_service_user_token option is requested, but the ksa\nrequest to load the service auth returns None.\n\nNote that today this only happens if the conf section is missing the\nauth_type option.  But this is a pretty good first indicator that the\nadmin forgot to populate auth options in general.\n\n[1] http://paste.openstack.org/show/623721/\n\nCloses-Bug: #1724689\nChange-Id: Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c90079a8890c7f294b654bb1d15fd3de9f2322f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c90079a8890c7f294b654bb1d15fd3de9f2322f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bb113796e1195d358c51cb2c4618c070feaed69f":{"kind":"REWORK","_number":4,"created":"2017-10-19 16:18:23.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/29/512329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/29/512329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/29/512329/4"}}},"commit":{"parents":[{"commit":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","subject":"Merge \"Refactor init_instance:resume_guests_state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1148a2d67b70dd767f798e1dfe0b4c7634d2f90c"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-19 16:18:22.000000000","tz":-300},"subject":"Service user token requested with no auth","message":"Service user token requested with no auth\n\nIf [service_user]send_service_user_token is requested, but no auth\ninformation is provided in the [service_user] section, the resulting\nerror is cryptic [1] and doesn\u0027t help the user understand what went\nwrong.\n\nWith this change, we raise an exception with a helpful message if the\n[service_user]send_service_user_token option is requested, but the ksa\nrequest to load the service auth returns None.\n\nNote that today this only happens if the conf section is missing the\nauth_type option.  But this is a pretty good first indicator that the\nadmin forgot to populate auth options in general.\n\n[1] http://paste.openstack.org/show/623721/\n\nCloses-Bug: #1724689\nChange-Id: Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb113796e1195d358c51cb2c4618c070feaed69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb113796e1195d358c51cb2c4618c070feaed69f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"cdaee2e114e4db886ea6eb1964df50020289e0fa":{"kind":"REWORK","_number":5,"created":"2017-10-19 16:21:29.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/29/512329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/29/512329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/29/512329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/29/512329/5"}}},"commit":{"parents":[{"commit":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","subject":"Merge \"Refactor init_instance:resume_guests_state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1148a2d67b70dd767f798e1dfe0b4c7634d2f90c"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-16 15:21:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-10-19 16:21:29.000000000","tz":-300},"subject":"Service user token requested with no auth","message":"Service user token requested with no auth\n\nIf [service_user]send_service_user_token is requested, but no auth\ninformation is provided in the [service_user] section, the resulting\nerror is cryptic [1] and doesn\u0027t help the user understand what went\nwrong.\n\nWith this change, we raise an exception with a helpful message if the\n[service_user]send_service_user_token option is requested, but the ksa\nrequest to load the service auth returns None.\n\nNote that today this only happens if the conf section is missing the\nauth_type option.  But this is a pretty good first indicator that the\nadmin forgot to populate auth options in general.\n\n[1] http://paste.openstack.org/show/623721/\n\nCloses-Bug: #1724689\nChange-Id: Ife1d26fd2373a02fc4f7f1bbffa7ec052457f11e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdaee2e114e4db886ea6eb1964df50020289e0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdaee2e114e4db886ea6eb1964df50020289e0fa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1148a2d67b70dd767f798e1dfe0b4c7634d2f90c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
