)]}'
{"id":"openstack%2Fnova~518382","triplet_id":"openstack%2Fnova~master~Ifc5f0b11e54d937d960d5dd9102e6287ee7320d8","project":"openstack/nova","branch":"master","topic":"bug/1731072","hashtags":[],"change_id":"Ifc5f0b11e54d937d960d5dd9102e6287ee7320d8","subject":"Test allocation_candidates with only sharing RPs","status":"MERGED","created":"2017-11-07 19:09:12.000000000","updated":"2017-11-13 16:28:17.000000000","submitted":"2017-11-13 16:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"518382-1510590497130-b5c247cf","meta_rev_id":"79875921545441f4b8f6e757083de6dc53a60dd2","_number":518382,"virtual_id_number":518382,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-11-10 17:23:59.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-11-10 17:27:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-11-10 17:24:01.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2017-11-10 16:17:17.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-11-10 16:29:31.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"date":"2017-11-10 17:27:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2017-11-13 16:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2017-11-10 15:09:17.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":2,"date":"2017-11-13 15:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":1,"date":"2017-11-13 15:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-08 13:52:41.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2017-11-10 14:37:57.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2017-11-10 15:46:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-10 16:17:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-10 17:23:59.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-10 17:24:01.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-11-10 17:27:42.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-11-10 17:27:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-11-13 15:03:13.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2017-11-13 16:28:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe2c3ff90c35e8bb335e65c28ea31b3a6e1b30be","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-07 19:09:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ea83faacc4a57aa3a3db400ba608623312a3291","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-07 20:02:56.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8432 : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f55e64196fbec4cde3c13cd8745cdebb7140d23b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-07 20:21:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/1/check/openstack-tox-pep8/2a4e72c/ : SUCCESS in 12m 16s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/1/check/openstack-tox-py27/909e0ab/ : SUCCESS in 10m 25s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/1/check/openstack-tox-py35/7df5b34/ : SUCCESS in 11m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/1/check/build-openstack-sphinx-docs/249c8f9/html/ : SUCCESS in 5m 47s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/1/check/legacy-tempest-dsvm-py35/e885264/ : SUCCESS in 1h 09m 21s\n- openstack-tox-functional http://logs.openstack.org/82/518382/1/check/openstack-tox-functional/63c42cd/ : SUCCESS in 20m 19s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/1/check/openstack-tox-functional-py35/db7cd52/ : SUCCESS in 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"67cca0bc7dc3360f500f65c32745082ac229fefe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-11-07 20:21:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"267e1c49ddb1a986661fe76ac3c1a121a881ec9e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-07 20:43:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/518382/1/check/nova-out-of-tree-pvm/627ac75 : SUCCESS in 1h 23m 59s\n- nova-in-tree-pvm http://184.172.12.213/82/518382/1/check/nova-in-tree-pvm/dbf554f : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d297acf1034c7b542daf113ff0fe717a79958c53","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-11-07 20:56:06.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-518382-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"0216ab2daf05d2c28920eb9e66f0c5511c75c5d6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-11-07 21:23:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/518382/1/ : SUCCESS in 2h 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9a275376f778af808e599a358c42e577f930c808","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2017-11-08 13:52:41.000000000","message":"Patch Set 1:\n\nrun nova-quobyteci-dsvm-volume","accounts_in_message":[],"_revision_number":1},{"id":"214b595991fd73924873664a4138c50628e3d873","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-11-08 17:21:12.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-518382-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4da43a9dea789f25e159191b23c1dfcf4ce2331","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-09 22:32:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2412c3d0a97fd15d1d3bfc2ea1c0b52d646bad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-10 00:00:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/2/check/openstack-tox-pep8/527faf5/ : SUCCESS in 7m 48s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/2/check/openstack-tox-py27/e99b9d0/ : SUCCESS in 25m 27s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-py35/e8adb41/ : SUCCESS in 12m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/2/check/build-openstack-sphinx-docs/924d7ec/html/ : SUCCESS in 6m 08s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/2/check/legacy-tempest-dsvm-py35/630e2ae/ : SUCCESS in 1h 25m 35s\n- openstack-tox-functional http://logs.openstack.org/82/518382/2/check/openstack-tox-functional/4ea36d5/ : SUCCESS in 22m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-functional-py35/86baf35/ : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e2ebabc30fc1cadca047daec24d99672923deb88","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-10 00:17:46.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8489 : SUCCESS in 1h 44m 44s","accounts_in_message":[],"_revision_number":2},{"id":"905c45e4be91737f69322f7187ebc55c152c1082","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-10 00:22:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/518382/2/check/nova-out-of-tree-pvm/81c6738 : SUCCESS in 1h 12m 52s\n- nova-in-tree-pvm http://184.172.12.213/82/518382/2/check/nova-in-tree-pvm/1965bf6 : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6fcb114de64efd7a882e65dbc5dde5724232fe30","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-11-10 00:49:07.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-518382-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f05070e4f2b8382f4de1f844d2f61c22ef57e3d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-11-10 01:32:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/518382/2/ : SUCCESS in 2h 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2d54eec6712e1cc9272009e9c855d4714c2c43aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-11-10 01:41:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/82/518382/2/check-vote/ext-nova-zuul/24ca8cb : FAILURE in 1h 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"647dba87a9c9db433ee0f42e73209db153a9af44","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-11-10 07:26:37.000000000","message":"Patch Set 2:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":2},{"id":"d3f02ebcd9df77275ed77906f12d52fb4d6f6719","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-10 08:40:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/2/check/openstack-tox-pep8/0faab8c/ : SUCCESS in 8m 06s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/2/check/openstack-tox-py27/cf2b9b2/ : SUCCESS in 12m 19s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-py35/092b9ad/ : SUCCESS in 35m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/2/check/build-openstack-sphinx-docs/5215b46/html/ : SUCCESS in 4m 39s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/2/check/legacy-tempest-dsvm-py35/059fb2f/ : SUCCESS in 1h 11m 43s\n- openstack-tox-functional http://logs.openstack.org/82/518382/2/check/openstack-tox-functional/1a95aff/ : POST_FAILURE in 31m 10s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-functional-py35/233c65e/ : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1b3605c434aaf2be8407bf0758ac920e747855fb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-11-10 09:03:09.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/518382/2/check-vote/ext-nova-zuul/fc70314 : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2d47e854b4c969164a8865b6ff968b69d117c09a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-11-10 10:52:16.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/518382/2/ : FAILURE in 3h 24m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b6a7876fdad1bc107335c67bffe7ff3201e1f171","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-10 13:12:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7afd3c7133cc1d5eded640867d88c61a673a61f0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-10 14:06:57.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/518382/2/check/nova-out-of-tree-pvm/e957565 : SUCCESS in 54m 13s\n- nova-in-tree-pvm http://184.172.12.213/82/518382/2/check/nova-in-tree-pvm/346edde : FAILURE in 47m 00s","accounts_in_message":[],"_revision_number":2},{"id":"76e26560c70084b92ac66f305c062c8d98812cfb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-11-10 14:12:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/82/518382/2/check-vote/ext-nova-zuul/14d2144 : FAILURE in 56m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5990c3fb89ff642eb9781a34c28582c8c42ac03c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-10 14:15:59.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8527 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3d3c098f8f98a876d6a59e15b3080bf6773cee70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-10 14:32:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/2/check/openstack-tox-pep8/b33d4b1/ : SUCCESS in 8m 02s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/2/check/openstack-tox-py27/98aebe3/ : SUCCESS in 13m 44s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-py35/30828f5/ : SUCCESS in 11m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/2/check/build-openstack-sphinx-docs/c638a3d/html/ : SUCCESS in 5m 32s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/2/check/legacy-tempest-dsvm-py35/bf59b2d/ : SUCCESS in 1h 10m 44s\n- openstack-tox-functional http://logs.openstack.org/82/518382/2/check/openstack-tox-functional/c0a6460/ : SUCCESS in 18m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/2/check/openstack-tox-functional-py35/1643241/ : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8617a54fb26672bb54ea569bc2ffccce01f3b047","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-11-10 14:32:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":2},{"id":"cb70ea19cc17015a644f0fa39bba2d278515f757","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-11-10 14:37:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\n++ excellent explanations in the code comments.","accounts_in_message":[],"_revision_number":2},{"id":"030d5304399b42284850e4d9e7ec62a131168b24","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-10 15:09:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"47428000a32e10f09c7d147e4be742f1406d468b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-11-10 15:46:44.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-518382-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"c0cad26947645b1032aae921e1bdfb0436893547","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-10 16:09:08.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/518382/3/check/nova-out-of-tree-pvm/02ed39a : SUCCESS in 58m 54s\n- nova-in-tree-pvm http://184.172.12.213/82/518382/3/check/nova-in-tree-pvm/71ce041 : FAILURE in 43m 58s","accounts_in_message":[],"_revision_number":3},{"id":"080de067570846e26186fa4e642bf0cc8bff7f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-10 16:16:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/3/check/openstack-tox-pep8/a7185a8/ : SUCCESS in 8m 45s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/3/check/openstack-tox-py27/3f94bca/ : SUCCESS in 13m 18s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/3/check/openstack-tox-py35/20e9714/ : SUCCESS in 12m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/3/check/build-openstack-sphinx-docs/4072365/html/ : SUCCESS in 4m 59s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/3/check/legacy-tempest-dsvm-py35/217c29a/ : SUCCESS in 1h 05m 13s\n- openstack-tox-functional http://logs.openstack.org/82/518382/3/check/openstack-tox-functional/2370309/ : SUCCESS in 18m 35s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/3/check/openstack-tox-functional-py35/832302b/ : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7158ead22d10a1fda58e4b2cdbaf90ed7a4f5649","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2017-11-10 16:17:17.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO_NOVA http://publiclogs.emc.com/scaleio/EMC_ScaleIO_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"8f6c1a5060a45a0754fc2b3a777814197e4b356a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-10 16:27:07.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8532 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":3},{"id":"11a1da1247076dacc1d46963a95d655103c5ca89","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2017-11-10 16:29:31.000000000","message":"Patch Set 3:\n\npowervm: recheck","accounts_in_message":[],"_revision_number":3},{"id":"49991813113af584b49b9eb9151f4f6cd06471bc","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-11-10 17:23:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/518382/3/check/nova-out-of-tree-pvm/acd8a7b : SUCCESS in 53m 33s\n- nova-in-tree-pvm http://184.172.12.213/82/518382/3/check/nova-in-tree-pvm/f0f3763 : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":3},{"id":"758599e0e7d5cec8d2c97f1adce18962d66c907f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-11-10 17:24:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/518382/3/ : SUCCESS in 2h 09m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f6108737776457e3000f1c3f7c6f50a8fc8269a7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-11-10 17:27:42.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-8536 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e86c325c99c46ed00ffaea21408c2f5def2076ab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-11-10 17:27:56.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/518382/3/check-vote/ext-nova-zuul/8624676 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ad09ab70975def95660746c7046c66009fc4b7d9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2017-11-13 15:03:13.000000000","message":"Patch Set 3: Workflow+1 Code-Review+2\n\nLooks good!\ncheers,\ngibi","accounts_in_message":[],"_revision_number":3},{"id":"0bb6aa4d1328dab6452748e6ae12b03ef14e4129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-13 15:03:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1406414663859adae2c54b91c2ceda6b0a72f4bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-13 16:28:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/82/518382/3/gate/openstack-tox-pep8/ec695ca/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/82/518382/3/gate/openstack-tox-py27/a76d117/ : SUCCESS in 10m 18s\n- openstack-tox-py35 http://logs.openstack.org/82/518382/3/gate/openstack-tox-py35/1cbbec4/ : SUCCESS in 11m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/82/518382/3/gate/build-openstack-sphinx-docs/e458600/html/ : SUCCESS in 5m 58s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/82/518382/3/gate/legacy-tempest-dsvm-py35/3d33546/ : SUCCESS in 1h 14m 41s\n- openstack-tox-functional http://logs.openstack.org/82/518382/3/gate/openstack-tox-functional/611ba4d/ : SUCCESS in 23m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/82/518382/3/gate/openstack-tox-functional-py35/de17852/ : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d61da87c2037085ac0c92f893c7761d4aca03aa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-13 16:28:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b80e885e0e9685757dc8f9364729a6fbf9be32d2","revisions":{"1f3a6eea24c72d128f978c0659a0b9ecd90e34a6":{"kind":"REWORK","_number":1,"created":"2017-11-07 19:09:12.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/82/518382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/518382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/518382/1"}}},"commit":{"parents":[{"commit":"8f021e2dc64307eb8d2ab68b283e9378499d5644","subject":"Test alloc candidates with same RC in cn \u0026 shared","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f021e2dc64307eb8d2ab68b283e9378499d5644"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-07 18:14:33.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-07 19:07:12.000000000","tz":-360},"subject":"Test allocation_candidates with only sharing RPs","message":"Test allocation_candidates with only sharing RPs\n\nThis change set contains a functional test demonstrating broken-ness\nin AllocationCandidates.get_by_filters when all the providers in the\ndatabase are sharing providers.  Whereas it is unclear what the expected\nbehavior should ultimately be, it is clear that the current behavior\nwould be considered broken under any sane design.\n\nChange-Id: Ifc5f0b11e54d937d960d5dd9102e6287ee7320d8\nRelated-Bug: #1730730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f3a6eea24c72d128f978c0659a0b9ecd90e34a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f3a6eea24c72d128f978c0659a0b9ecd90e34a6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8f021e2dc64307eb8d2ab68b283e9378499d5644","is_merged_in_target_branch":false,"change_id":"I42edf102379cf329aa2252ab779a9f945f5fc155","change_number":513149,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"17ba090e2561aa251f79b0b3746409e51414bfe2":{"kind":"REWORK","_number":2,"created":"2017-11-09 22:32:18.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/82/518382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/518382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/518382/2"}}},"commit":{"parents":[{"commit":"9c14daa4c04a303f99ec57e7f7ae46771f6f6304","subject":"Test alloc candidates with same RC in cn \u0026 shared","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c14daa4c04a303f99ec57e7f7ae46771f6f6304"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-07 18:14:33.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-09 22:15:32.000000000","tz":-360},"subject":"Test allocation_candidates with only sharing RPs","message":"Test allocation_candidates with only sharing RPs\n\nThis change set contains a functional test demonstrating broken-ness\nin AllocationCandidates.get_by_filters when all the providers in the\ndatabase are sharing providers.\n\nChange-Id: Ifc5f0b11e54d937d960d5dd9102e6287ee7320d8\nRelated-Bug: #1730730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17ba090e2561aa251f79b0b3746409e51414bfe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17ba090e2561aa251f79b0b3746409e51414bfe2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9c14daa4c04a303f99ec57e7f7ae46771f6f6304","is_merged_in_target_branch":false,"change_id":"I42edf102379cf329aa2252ab779a9f945f5fc155","change_number":513149,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b80e885e0e9685757dc8f9364729a6fbf9be32d2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-11-10 15:09:17.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/82/518382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/518382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/518382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/518382/3"}}},"commit":{"parents":[{"commit":"d4398f715f098b9edbc0be8612b6c079a8e607af","subject":"Test alloc candidates with same RC in cn \u0026 shared","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4398f715f098b9edbc0be8612b6c079a8e607af"}]}],"author":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-07 18:14:33.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2017-11-10 15:09:10.000000000","tz":-360},"subject":"Test allocation_candidates with only sharing RPs","message":"Test allocation_candidates with only sharing RPs\n\nThis change set contains a functional test demonstrating broken-ness\nin AllocationCandidates.get_by_filters when all the providers in the\ndatabase are sharing providers.\n\nChange-Id: Ifc5f0b11e54d937d960d5dd9102e6287ee7320d8\nRelated-Bug: #1730730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b80e885e0e9685757dc8f9364729a6fbf9be32d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b80e885e0e9685757dc8f9364729a6fbf9be32d2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d4398f715f098b9edbc0be8612b6c079a8e607af","is_merged_in_target_branch":true,"change_id":"I42edf102379cf329aa2252ab779a9f945f5fc155","change_number":513149,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
