)]}'
{"id":"openstack%2Fnova~526823","triplet_id":"openstack%2Fnova~master~I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614","project":"openstack/nova","branch":"master","topic":"bug/1734625","hashtags":[],"change_id":"I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614","subject":"[placement] Add sending global request ID in post","status":"MERGED","created":"2017-12-09 06:26:47.000000000","updated":"2018-02-06 04:51:51.000000000","submitted":"2018-02-06 04:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"526823-1517889941342-897524af","meta_rev_id":"e17fe60ae431fec98404c84fec47125365fdd9d2","_number":526823,"virtual_id_number":526823,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-02-06 01:14:10.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},{"value":0,"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},{"value":0,"date":"2018-02-06 02:23:48.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-05 01:26:46.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2018-02-05 00:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"date":"2018-02-06 00:56:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-02-06 04:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2018-02-05 05:41:02.000000000","_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-05 02:17:11.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-05 23:57:41.000000000","_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"date":"2018-02-06 03:33:03.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},{"value":0,"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2018-02-05 16:39:28.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2018-02-04 23:40:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2018-02-05 12:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},{"value":0,"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2018-02-05 16:39:28.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},{"value":0,"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},{"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-28 02:11:10.000000000","updated_by":{"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},"reviewer":{"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},"state":"REVIEWER"},{"updated":"2017-12-28 03:10:09.000000000","updated_by":{"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},"reviewer":{"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},"state":"REVIEWER"},{"updated":"2018-01-02 15:09:37.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-11 12:50:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-22 06:24:30.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2018-01-22 14:48:08.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-01-23 07:12:34.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2018-01-23 08:01:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-02 01:09:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-02 08:07:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-02-05 00:40:59.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-05 01:26:46.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-05 02:17:11.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-05 05:41:02.000000000","updated_by":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"reviewer":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"state":"REVIEWER"},{"updated":"2018-02-05 12:33:56.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-02-05 16:39:28.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-02-06 00:56:34.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-02-06 01:14:10.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-06 02:23:48.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-02-06 03:33:03.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-02-06 04:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d17c0b4915e214692e77f4f0b8dfaf7e1de29e50","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-09 06:26:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d611aff14b2487a8af10d2cb2e00d1143eae2944","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 06:27:20.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"241b89dd2b5a12c8bbb7f288a6dd1c37be4b106e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-09 06:27:58.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5176989d6e29835b79468a340a43000ef01c2bba","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-09 06:28:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6bb052d0e5cc8719aab1ae952e46db2fc2e6a164","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-09 06:28:48.000000000","message":"Patch Set 1:\n\nMerge failed. To try again use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems","accounts_in_message":[],"_revision_number":1},{"id":"3d3817253ec1189075a4182fcb99cfbc09eeb2f9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-09 08:23:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71709c291220503365683a4e674c4648275e40e3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-09 08:40:45.000000000","message":"Patch Set 2: Verified-1\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/36f805b : FAILURE in 16m 16s","accounts_in_message":[],"_revision_number":2},{"id":"18884f0df2bcae48f81945e465c3ac63d393bf35","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-09 09:17:29.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10344 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2e84172595ca9a214f5df796ac86671801ee5814","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-09 09:19:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/2/check/check-dsvm-tempest-vz7-exe-minimal/745e893 : SUCCESS in 51m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"ad8cad4819e435d382f9e58300847de6ee8a36f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-09 09:21:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/tempest-dsvm-full-xenial/913f631/ : SUCCESS in 55m 57s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/grenade-dsvm-xenial/f186115/ : SUCCESS in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d093d3a5a48e2b969e5aae7bc7b5ef48c9a73073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 09:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/2/check/openstack-tox-pep8/4b5b793/ : SUCCESS in 8m 25s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/2/check/openstack-tox-py27/263b522/ : SUCCESS in 9m 48s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-py35/b891cd8/ : SUCCESS in 21m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/2/check/build-openstack-sphinx-docs/812b0c4/html/ : SUCCESS in 6m 12s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-py35/9d19c60/ : SUCCESS in 1h 08m 12s\n- openstack-tox-functional http://logs.openstack.org/23/526823/2/check/openstack-tox-functional/29beca8/ : SUCCESS in 22m 29s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-functional-py35/b3c4f5e/ : SUCCESS in 17m 51s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode/a55f00e/ : SUCCESS in 58m 52s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/0ba5c13/ : SUCCESS in 51m 14s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-cells/91e60cd/ : SUCCESS in 46m 15s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/9afbcab/ : FAILURE in 1h 22m 36s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-multinode-full/7fb5128/ : SUCCESS in 1h 30m 28s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-nova-next-full/74d774e/ : SUCCESS in 1h 24m 22s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-multinode-live-migration/d22bc5f/ : SUCCESS in 38m 10s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1569d07/ : SUCCESS in 48m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-full/2d7a302/ : SUCCESS in 1h 26m 50s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron/925ff01/ : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2fae867e0377afd93b13a2be58e0781a1b94e81e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-09 10:14:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/2/check/dsvm-tempest-neutron-network/9d1af6d : SUCCESS in 1h 49m 39s","accounts_in_message":[],"_revision_number":2},{"id":"43ac3911eadbe495c2d05c8477981a2f5c310386","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-09 10:24:10.000000000","message":"Patch Set 2:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":2},{"id":"5820ae90f5cd848be17d98a745dc4faf567af672","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-09 10:36:56.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/2/check/dsvm-tempest-xen/b6a5595 : FAILURE in 2h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7b71cdbaa4f605e44f1c028a1214727ee1aa267c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-09 10:44:12.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/98bb174 : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f8f04e9ecd3465ec91a78b56c9234ae786a2d68f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-09 10:48:15.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"35194ec1962ef6383a648f18474e813cd1b4181e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-09 10:49:18.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/2/check-vote/ext-nova-zuul/c6a916e : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":2},{"id":"cbec9cceea69b40f1eae3e1cf120146f4a296f52","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-09 11:27:37.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/tempest-dsvm-full-xenial/73dc744/ : SUCCESS in 1h 02m 15s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/grenade-dsvm-xenial/540b7c5/ : SUCCESS in 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4309f0023250bbdb92d37d8d7bf9727f06fb0fd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 12:00:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/2/check/openstack-tox-pep8/21ef6c2/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/2/check/openstack-tox-py27/f3ebf47/ : SUCCESS in 12m 18s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-py35/08f6f22/ : SUCCESS in 10m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/2/check/build-openstack-sphinx-docs/57781bd/html/ : SUCCESS in 6m 09s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-py35/f6b4504/ : SUCCESS in 1h 15m 18s\n- openstack-tox-functional http://logs.openstack.org/23/526823/2/check/openstack-tox-functional/69da293/ : SUCCESS in 18m 15s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-functional-py35/4ab025f/ : SUCCESS in 22m 34s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode/f7ccd7e/ : SUCCESS in 53m 59s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b6b7e1b/ : SUCCESS in 1h 02m 51s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-cells/bc04931/ : SUCCESS in 37m 16s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3790820/ : FAILURE in 1h 05m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-multinode-full/a25a215/ : SUCCESS in 1h 34m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-nova-next-full/51b2964/ : SUCCESS in 1h 14m 46s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-multinode-live-migration/2a9c618/ : SUCCESS in 33m 02s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a785db3/ : SUCCESS in 38m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-full/e3192aa/ : FAILURE in 1h 23m 06s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron/633f999/ : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":2},{"id":"815e098da5cc274cf8088ec76e7597b642e3bedb","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-09 12:03:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8a3c28cd09e797773f0531f9009c6e3856851b2a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-09 12:57:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/2/check-vote/ext-nova-zuul/f00410c : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e4ac67de115bfcad37702e9dc6a135d2d776f51f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-09 13:02:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/tempest-dsvm-full-xenial/ffc7122/ : SUCCESS in 57m 25s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/2/check/grenade-dsvm-xenial/d32414f/ : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ec68d141ab457f96fdf5ab8fc883d419e7e67af","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-09 13:06:07.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10351 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"7da10d076216ad92c7577d086cd87ba707f2ff6d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-09 13:24:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/2/check/check-dsvm-tempest-vz7-exe-minimal/acbe4df : SUCCESS in 1h 05m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"d559d31507bfa5f4ca908246fe48816e5e9e8286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-09 14:07:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/2/check/openstack-tox-pep8/b6e5c7f/ : SUCCESS in 11m 26s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/2/check/openstack-tox-py27/1a6a253/ : SUCCESS in 16m 04s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-py35/25dd797/ : SUCCESS in 11m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/2/check/build-openstack-sphinx-docs/10b5f57/html/ : SUCCESS in 5m 46s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-py35/3adbf1d/ : SUCCESS in 1h 25m 29s\n- openstack-tox-functional http://logs.openstack.org/23/526823/2/check/openstack-tox-functional/9267b67/ : SUCCESS in 22m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/2/check/openstack-tox-functional-py35/b7efe68/ : SUCCESS in 22m 53s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode/e6cc3b1/ : SUCCESS in 54m 32s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b88f350/ : SUCCESS in 54m 36s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-cells/8b1b082/ : SUCCESS in 37m 00s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/c02516e/ : FAILURE in 1h 02m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-multinode-full/c74ec97/ : SUCCESS in 2h 01m 34s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-nova-next-full/79aae95/ : SUCCESS in 1h 19m 02s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-multinode-live-migration/a2d6014/ : SUCCESS in 48m 53s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/55ec7b6/ : SUCCESS in 40m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/2/check/legacy-tempest-dsvm-neutron-full/5dee749/ : SUCCESS in 1h 23m 24s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/2/check/legacy-grenade-dsvm-neutron/5b022ad/ : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d8abdf9fa286c07889d1c23b384595a486a3ce86","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-09 14:21:27.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/2/ : FAILURE in 3h 55m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a15a49c86ca8ac5326627ea44ecf7390bbd6653c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-09 14:53:57.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"28c96416d69702706781044e5899dbaf41ac13fd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-09 15:08:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/2/check-vote/ext-nova-zuul/5bed68f : FAILURE in 54m 35s","accounts_in_message":[],"_revision_number":2},{"id":"26a0fcc18dc66ef3e393281798709692477008a0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-09 22:54:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/2/check-nova/Nova-ML2-Sriov/40ffaa2 : SUCCESS in 1h 03m 27s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/2/check-nova/Nova-MACVTAP-ML2-Sriov/15677f2 : FAILURE in 1h 03m 33s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3a057ce554d51014047c9d3073f8405ada70386","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-11 00:50:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1624d3988841b371d7a3208e04b3a99260d9d7a8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-11 00:50:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4f09c8155ae1ca831f1e68643dee8bdfe22fa401","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-11 01:05:11.000000000","message":"Patch Set 4: Verified-1\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/40870a7 : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":4},{"id":"10465db7bb9a711b90c0fe7917d20fac85138b7c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-11 01:43:17.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10374 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":4},{"id":"488c9e91acfd5b8dca3b68404dbea72d085f272d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-11 01:46:53.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/4/check/nova-out-of-tree-pvm/6b4244e : FAILURE in 55m 45s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/4/check/nova-in-tree-pvm/31d3cb0 : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f293cf03e0eb15053e14dff905327c7ec4f8f477","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-11 02:03:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/4/check/tempest-dsvm-full-xenial/1a565bd/ : SUCCESS in 1h 09m 55s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/4/check/grenade-dsvm-xenial/983e540/ : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38cb3e06b7e9761fcdae7ccf80b42600cc8ae949","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-11 02:24:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/4/check/openstack-tox-pep8/8056034/ : SUCCESS in 7m 16s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/4/check/openstack-tox-py27/b564fc5/ : SUCCESS in 9m 58s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/4/check/openstack-tox-py35/312d636/ : SUCCESS in 11m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/4/check/build-openstack-sphinx-docs/817b98c/html/ : SUCCESS in 5m 40s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-py35/7cbd6ce/ : SUCCESS in 1h 18m 03s\n- openstack-tox-functional http://logs.openstack.org/23/526823/4/check/openstack-tox-functional/8a93e88/ : SUCCESS in 21m 38s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/4/check/openstack-tox-functional-py35/fec2e84/ : SUCCESS in 19m 58s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron-multinode/abe1b27/ : SUCCESS in 51m 56s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/94208da/ : SUCCESS in 57m 31s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-cells/d7ff5cf/ : SUCCESS in 37m 35s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/cb7048b/ : FAILURE in 1h 07m 08s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-multinode-full/7c05247/ : SUCCESS in 1h 30m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-nova-next-full/75ace37/ : SUCCESS in 1h 17m 07s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-multinode-live-migration/8dedb67/ : SUCCESS in 38m 21s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6a512f0/ : SUCCESS in 38m 00s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-full/c625d14/ : SUCCESS in 1h 18m 11s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron/4a6b510/ : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":4},{"id":"92739b2d3700215306fa0499310d3899d94769af","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-11 02:57:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/4/check/dsvm-tempest-xen/43ae8e6 : SUCCESS in 2h 01m 21s","accounts_in_message":[],"_revision_number":4},{"id":"8c0e642402233db059bcec2c20ba7af9077358aa","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-11 02:59:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/4/check/check-dsvm-tempest-vz7-exe-minimal/2b5612c : SUCCESS in 55m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"6c15196828145d26544245023bda41ecdda24288","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-11 03:18:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/4/check/dsvm-tempest-neutron-network/82443f5 : SUCCESS in 1h 55m 45s","accounts_in_message":[],"_revision_number":4},{"id":"433888c1cfd58efcae677d02b8b10146251b9edc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-11 03:27:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/4/check-vote/ext-nova-zuul/bf5638b : FAILURE in 1h 01m 29s","accounts_in_message":[],"_revision_number":4},{"id":"dccc541e2571652380c5ca7c2865613456078608","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-11 03:38:20.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"35efbb177dc9ecb4f545b1bcbbee0b3304487331","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-11 05:05:17.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/4/ : FAILURE in 2h 54m 55s","accounts_in_message":[],"_revision_number":4},{"id":"bb3b7be222c49940e16bfab5f465cc26c22575cb","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-11 05:53:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/4/check-nova/Nova-ML2-Sriov/b16d572 : FAILURE in 49m 54s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/4/check-nova/Nova-MACVTAP-ML2-Sriov/e5c0c7c : FAILURE in 1h 21m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"2c6738eb4f9ee4670fea450c3b0f5d44acc7a0bb","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2017-12-11 20:16:42.000000000","message":"Patch Set 4:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":4},{"id":"93e287495ec384c2e61f5ac02c9cd1b23cc7ac18","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-11 21:41:19.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/4/check/tempest-dsvm-full-xenial/11371e8/ : SUCCESS in 1h 10m 59s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/4/check/grenade-dsvm-xenial/501b3e4/ : SUCCESS in 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93f33d831abc3329aebc557e6edb133d3273f848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-11 22:29:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/4/check/openstack-tox-pep8/ff38b71/ : SUCCESS in 8m 45s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/4/check/openstack-tox-py27/264c74b/ : SUCCESS in 14m 06s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/4/check/openstack-tox-py35/8a55a73/ : SUCCESS in 13m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/4/check/build-openstack-sphinx-docs/4f6d44a/html/ : SUCCESS in 5m 45s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-py35/0117eb2/ : SUCCESS in 1h 20m 52s\n- openstack-tox-functional http://logs.openstack.org/23/526823/4/check/openstack-tox-functional/1c20c50/ : SUCCESS in 21m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/4/check/openstack-tox-functional-py35/b77579b/ : SUCCESS in 22m 19s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron-multinode/9ad1ee3/ : SUCCESS in 54m 40s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/28e7205/ : SUCCESS in 51m 52s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-cells/1131085/ : SUCCESS in 38m 02s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/af377dc/ : FAILURE in 1h 10m 35s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-multinode-full/302e1f0/ : SUCCESS in 1h 41m 54s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-nova-next-full/5af920d/ : SUCCESS in 1h 12m 19s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-multinode-live-migration/31324f3/ : SUCCESS in 36m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9189494/ : SUCCESS in 40m 59s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/4/check/legacy-tempest-dsvm-neutron-full/3106b5c/ : SUCCESS in 1h 15m 52s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/4/check/legacy-grenade-dsvm-neutron/5b9b33b/ : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":4},{"id":"9ea2ba81ee7cae9fb1062f6d0cce1cba75d05e90","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-12 01:06:35.000000000","message":"Patch Set 4: -Verified\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7fe5518 : SUCCESS in 1h 45m 55s","accounts_in_message":[],"_revision_number":4},{"id":"ce73d645142ef219a0f75d2e8f00852cbab1875f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-12 04:35:11.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/4/check-vote/ext-nova-zuul/08c5c0c : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":4},{"id":"61dd0aa7941e42cbab6acd8a512464d66757a06d","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-12 06:10:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f857e4306f9898c972cc8904fb78f491dd7d5ade","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-12 07:01:23.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10494 : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":5},{"id":"88554cb9bff4981a0074e83170d99f12552d328e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-12 07:07:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/5/check/tempest-dsvm-full-xenial/74f1fc2/ : SUCCESS in 54m 33s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/5/check/grenade-dsvm-xenial/48d0668/ : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d766bde122e3b22e156751ed30702b58e2b52401","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-12 07:07:33.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/5/check/nova-out-of-tree-pvm/7059065 : FAILURE in 56m 24s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/5/check/nova-in-tree-pvm/377c42e : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":5},{"id":"ada3ce70035c831f15a8a03fcf1eb33c271f48c9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-12 08:04:17.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/18aee97 : SUCCESS in 1h 51m 33s","accounts_in_message":[],"_revision_number":5},{"id":"a33d0858abaee8cdc5715c4e47364e6308c92174","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-12 08:24:15.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/5/check/dsvm-tempest-xen/02c7627 : FAILURE in 2h 06m 19s","accounts_in_message":[],"_revision_number":5},{"id":"14ca3ee208e3f04547a62c33d40a48f2889189de","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-12 09:00:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/5/check/check-dsvm-tempest-vz7-exe-minimal/cf4b404 : SUCCESS in 49m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"58ad673dd567e9c74b416765c75b89a729886353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-12 09:21:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/5/check/openstack-tox-pep8/5b31107/ : SUCCESS in 8m 10s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/5/check/openstack-tox-py27/9586a55/ : TIMED_OUT in 50m 01s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/5/check/openstack-tox-py35/c1ad93b/ : POST_FAILURE in 55m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/5/check/build-openstack-sphinx-docs/958dbdd/html/ : SUCCESS in 10m 27s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-py35/bc32b5d/ : POST_FAILURE in 2h 10m 48s\n- openstack-tox-functional http://logs.openstack.org/23/526823/5/check/openstack-tox-functional/9e47ae1/ : TIMED_OUT in 49m 08s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/5/check/openstack-tox-functional-py35/6d7a45a/ : TIMED_OUT in 51m 38s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/5/check/legacy-grenade-dsvm-neutron-multinode/3e40da8/ : POST_FAILURE in 2h 08m 43s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/00f1f0c/ : POST_FAILURE in 2h 19m 11s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-cells/9a1d28d/ : POST_FAILURE in 2h 03m 46s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/c9fc8c1/ : POST_FAILURE in 2h 17m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-neutron-multinode-full/453eb77/ : POST_FAILURE in 2h 17m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-neutron-nova-next-full/161e801/ : POST_FAILURE in 2h 09m 43s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-multinode-live-migration/7f445aa/ : POST_FAILURE in 2h 18m 02s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/42988bd/ : TIMED_OUT in 1h 37m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/5/check/legacy-tempest-dsvm-neutron-full/3c04db7/ : POST_FAILURE in 2h 18m 08s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/5/check/legacy-grenade-dsvm-neutron/1bd0952/ : POST_FAILURE in 2h 16m 50s","accounts_in_message":[],"_revision_number":5},{"id":"28a668959559329ba41ccc4af13117219dbb366e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-12 09:28:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/5/check/dsvm-tempest-neutron-network/89ad075 : SUCCESS in 3h 13m 04s","accounts_in_message":[],"_revision_number":5},{"id":"bd267579d30876c3edc3ae1a93c93a93d0fe9e6b","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-12 13:07:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/5/check-nova/Nova-ML2-Sriov/6a337cd : SUCCESS in 1h 35m 47s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/5/check-nova/Nova-MACVTAP-ML2-Sriov/ce3fb48 : FAILURE in 1h 14m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"8890814bf30d062f1a5f277143d27f24c975ff6a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-13 02:06:19.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"142897856b79f6981045178bb53d72df7715e052","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-13 03:01:17.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10577 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3da621ce23fa0999bcacb67d2fa6b5527ded7f28","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-13 03:03:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/6/check/tempest-dsvm-full-xenial/2267455/ : SUCCESS in 53m 55s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/6/check/grenade-dsvm-xenial/83324fe/ : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d3a413725d82cf76a6fd0f0ed122f5c1b5dba6a0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-13 03:04:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/6/check/nova-out-of-tree-pvm/7fd8bf6 : SUCCESS in 57m 12s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/6/check/nova-in-tree-pvm/d604ed8 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":6},{"id":"0777c8ac9444be34d23933b089429bd87f8604b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 03:52:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/6/check/openstack-tox-pep8/267b57c/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/6/check/openstack-tox-py27/989d02d/ : SUCCESS in 13m 18s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/6/check/openstack-tox-py35/b7156d1/ : SUCCESS in 13m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/6/check/build-openstack-sphinx-docs/ade49f6/html/ : SUCCESS in 5m 17s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-py35/e322bd2/ : SUCCESS in 1h 14m 49s\n- openstack-tox-functional http://logs.openstack.org/23/526823/6/check/openstack-tox-functional/fa4375c/ : SUCCESS in 24m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/6/check/openstack-tox-functional-py35/b4f95a1/ : FAILURE in 20m 54s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/6/check/legacy-grenade-dsvm-neutron-multinode/1cca74a/ : SUCCESS in 1h 00m 12s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/6/check/legacy-grenade-dsvm-neutron-multinode-live-migration/fc10192/ : SUCCESS in 52m 51s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-cells/ce7a1b7/ : SUCCESS in 38m 01s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/d469f28/ : SUCCESS in 1h 01m 39s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-neutron-multinode-full/e808698/ : SUCCESS in 1h 32m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-neutron-nova-next-full/f02f083/ : SUCCESS in 1h 16m 54s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-multinode-live-migration/598c0e8/ : SUCCESS in 38m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/6/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9deca3c/ : SUCCESS in 38m 19s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/6/check/legacy-tempest-dsvm-neutron-full/c36c2a5/ : SUCCESS in 1h 23m 15s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/6/check/legacy-grenade-dsvm-neutron/c0d208b/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":6},{"id":"2e567c112ab21062625b80008ec7b822cf5cf73d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-13 04:11:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/6/check/dsvm-tempest-neutron-network/3e18a44 : SUCCESS in 1h 55m 08s","accounts_in_message":[],"_revision_number":6},{"id":"fe33c07c07b2870a0429066fa63fa0e23fc481fa","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-13 04:19:48.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/6/check/dsvm-tempest-xen/5418457 : FAILURE in 2h 03m 21s","accounts_in_message":[],"_revision_number":6},{"id":"1930dfbd3a4aa4019b3b109e044288649488e718","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-13 04:22:32.000000000","message":"Patch Set 6: Verified-1\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/4448b21 : FAILURE in 2h 13m 32s","accounts_in_message":[],"_revision_number":6},{"id":"78a5bad5236ba692cc9330bf3b993bc5bdf9cb40","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-13 08:29:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/6/ : FAILURE in 3h 55m 08s","accounts_in_message":[],"_revision_number":6},{"id":"e6a6d88451a29627797749fc5181acab98537871","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-14 09:17:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"66d25475b06dacf880d75c9a67fce56a2d0bda53","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-14 10:01:58.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10730 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":7},{"id":"223dd4f3be24b6c5d952949546da041b9789194c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-14 10:05:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/7/check/nova-out-of-tree-pvm/9ef35dd : SUCCESS in 46m 24s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/7/check/nova-in-tree-pvm/9276f22 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":7},{"id":"9c9b9857669c205059d52c381bb9b1229af4f339","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-14 10:15:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/7/check/tempest-dsvm-full-xenial/0562f73/ : SUCCESS in 56m 50s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/7/check/grenade-dsvm-xenial/9d6a19f/ : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c8ed67c30221aaf38f98c01b1fea8e44b12ac505","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-14 10:44:16.000000000","message":"Patch Set 7: Verified+1\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1ecda63 : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":7},{"id":"8a3c58c3bda2cdc377911cfb8ee2febe9edac2e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 11:10:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/7/check/openstack-tox-pep8/5ad8e21/ : SUCCESS in 8m 15s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/7/check/openstack-tox-py27/c52d082/ : SUCCESS in 10m 58s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/7/check/openstack-tox-py35/ae953cf/ : SUCCESS in 12m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/7/check/build-openstack-sphinx-docs/22a075e/html/ : SUCCESS in 4m 35s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-py35/23d1c95/ : SUCCESS in 1h 21m 32s\n- openstack-tox-functional http://logs.openstack.org/23/526823/7/check/openstack-tox-functional/fe37813/ : SUCCESS in 21m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/7/check/openstack-tox-functional-py35/710a8f9/ : SUCCESS in 20m 08s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/7/check/legacy-grenade-dsvm-neutron-multinode/e41e9b4/ : SUCCESS in 54m 09s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/7/check/legacy-grenade-dsvm-neutron-multinode-live-migration/159cb91/ : SUCCESS in 52m 32s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-cells/88f32ef/ : SUCCESS in 35m 02s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6337c7b/ : SUCCESS in 1h 15m 00s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-neutron-multinode-full/7223fb2/ : SUCCESS in 1h 35m 53s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-neutron-nova-next-full/768ee16/ : SUCCESS in 1h 11m 43s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-multinode-live-migration/79f102c/ : SUCCESS in 35m 47s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/7/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/658e57b/ : SUCCESS in 41m 13s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/7/check/legacy-tempest-dsvm-neutron-full/0c32375/ : SUCCESS in 1h 23m 23s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/7/check/legacy-grenade-dsvm-neutron/30daa70/ : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":7},{"id":"25beb69171fa4f5891f1a386cc7e0190cd1840d7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-14 11:27:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/7/check-nova/Nova-ML2-Sriov/677536e : FAILURE in 1h 02m 59s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/7/check-nova/Nova-MACVTAP-ML2-Sriov/f9d10f5 : SUCCESS in 1h 15m 32s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"d1a70bb418630a1c3d68445ecc751367ecb2c4f5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-14 11:29:45.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"77e4de3182142186d7dc804b0f02f24dc8744700","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-14 11:33:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/7/check/dsvm-tempest-neutron-network/3a0348d : SUCCESS in 2h 03m 34s","accounts_in_message":[],"_revision_number":7},{"id":"0cd2041c6e2bb7f738b84773faca9d819fc7c7c7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-14 11:52:45.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/7/check/dsvm-tempest-xen/4417b96 : FAILURE in 2h 32m 47s","accounts_in_message":[],"_revision_number":7},{"id":"a61c57d565f64c97dfc3c6fa49bc3a0edeb45162","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-14 12:30:36.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/7/check-vote/ext-nova-zuul/229456d : FAILURE in 1h 07m 59s","accounts_in_message":[],"_revision_number":7},{"id":"03e7ff0da246b197063ca016b9c5cf934bfefeb8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-15 00:15:30.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"50125a41c1d02d922223ba0fef2458dc0a9caeb5","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-15 01:06:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/8/check/nova-out-of-tree-pvm/5951d86 : SUCCESS in 49m 43s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/8/check/nova-in-tree-pvm/e839826 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":8},{"id":"e20ac7e5979a2e99311930326b527adf86939fc3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-15 01:06:22.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-10871 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":8},{"id":"fd3e1688551b184c7f482f62543e75b0a66434a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-15 01:14:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/8/check/tempest-dsvm-full-xenial/7832957/ : SUCCESS in 56m 07s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/8/check/grenade-dsvm-xenial/a3111e6/ : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d64f3ab2dd34addd5c72a1489ee1c75c99e24907","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-15 01:52:11.000000000","message":"Patch Set 8:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/85a589f : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":8},{"id":"c185a9d335db1330544af0347c7041a5fa798dc1","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-15 02:28:45.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/8/check/dsvm-tempest-neutron-network/972bb7a : SUCCESS in 2h 01m 37s","accounts_in_message":[],"_revision_number":8},{"id":"59952d4dc10969307fd81c84c75e4ed0b70f06ac","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-15 02:39:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/8/check/dsvm-tempest-xen/042aaee : SUCCESS in 2h 13m 54s","accounts_in_message":[],"_revision_number":8},{"id":"1d5e3ef1666f315acdf1a54db0dbff41e41a8193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 02:41:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/8/check/openstack-tox-pep8/7ed53e3/ : SUCCESS in 8m 07s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/8/check/openstack-tox-py27/d98b872/ : SUCCESS in 14m 26s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/8/check/openstack-tox-py35/ed194b6/ : SUCCESS in 11m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/8/check/build-openstack-sphinx-docs/49118fe/html/ : SUCCESS in 6m 18s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-py35/6b501b0/ : SUCCESS in 1h 09m 35s\n- openstack-tox-functional http://logs.openstack.org/23/526823/8/check/openstack-tox-functional/b3916ba/ : SUCCESS in 23m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/8/check/openstack-tox-functional-py35/ee893f4/ : SUCCESS in 22m 09s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/8/check/legacy-grenade-dsvm-neutron-multinode/458bd98/ : SUCCESS in 57m 07s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/8/check/legacy-grenade-dsvm-neutron-multinode-live-migration/722e36d/ : SUCCESS in 1h 01m 08s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-cells/4e68556/ : SUCCESS in 35m 18s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/b2fe43b/ : SUCCESS in 1h 12m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-neutron-multinode-full/166f87f/ : SUCCESS in 1h 37m 15s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-neutron-nova-next-full/32ca000/ : SUCCESS in 1h 14m 21s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-multinode-live-migration/4ba7b94/ : SUCCESS in 40m 07s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/8/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/78fd3dc/ : SUCCESS in 39m 22s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/8/check/legacy-tempest-dsvm-neutron-full/651671c/ : SUCCESS in 1h 21m 32s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/8/check/legacy-grenade-dsvm-neutron/b1fc69e/ : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":8},{"id":"f8ac19ddcc44723a50ed703f6d1de8e683a45358","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-15 03:06:57.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"082d3192319b9600686dde318b17cb725bfd6ab3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-15 04:28:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/8/ : SUCCESS in 3h 13m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a63b2c98a5b819c2bd20e7a680d944a9b55945c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-15 07:36:08.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/8/check-vote/ext-nova-zuul/66b3d16 : FAILURE in 53m 03s","accounts_in_message":[],"_revision_number":8},{"id":"3a3c40f101ce4ce92394f0a972c33d1444a76627","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-20 22:05:40.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"bd810c7425560a43119e01d82194f264bd548e78","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-20 22:50:59.000000000","message":"Patch Set 9:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/9/check/nova-out-of-tree-pvm/2f3e971 : SUCCESS in 44m 34s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/9/check/nova-in-tree-pvm/cfa9cfa : FAILURE in 34m 42s","accounts_in_message":[],"_revision_number":9},{"id":"96d028d3f0789e4e95c5f1b3e13baa46713f06c2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-20 23:04:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/9/check/tempest-dsvm-full-xenial/c192c0a/ : SUCCESS in 56m 22s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/9/check/grenade-dsvm-xenial/4b9262e/ : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"18fafbee5b6b0422dcf95bf5878ac8ff90f5da71","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-20 23:07:36.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11229 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":9},{"id":"882bed8bd017a09dc5448d70592d88f7edb796b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-20 23:56:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/9/check/openstack-tox-pep8/31174c9/ : SUCCESS in 8m 49s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/9/check/openstack-tox-py27/36c1344/ : SUCCESS in 11m 30s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/9/check/openstack-tox-py35/f84210a/ : SUCCESS in 12m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/9/check/build-openstack-sphinx-docs/5d2f726/html/ : SUCCESS in 5m 27s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-py35/aa1ca73/ : SUCCESS in 1h 25m 35s\n- openstack-tox-functional http://logs.openstack.org/23/526823/9/check/openstack-tox-functional/650594a/ : SUCCESS in 22m 04s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/9/check/openstack-tox-functional-py35/78e53b0/ : SUCCESS in 19m 16s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/9/check/legacy-grenade-dsvm-neutron-multinode/ed485f7/ : SUCCESS in 55m 13s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/9/check/legacy-grenade-dsvm-neutron-multinode-live-migration/132255b/ : SUCCESS in 53m 54s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-cells/25103cd/ : SUCCESS in 46m 53s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/a2ffcbd/ : SUCCESS in 56m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-neutron-multinode-full/fde3219/ : SUCCESS in 1h 32m 51s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-neutron-nova-next-full/905d9a2/ : SUCCESS in 1h 15m 37s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-multinode-live-migration/6a3ab39/ : SUCCESS in 38m 40s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/9/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/762c8b9/ : SUCCESS in 43m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/9/check/legacy-tempest-dsvm-neutron-full/24111e8/ : SUCCESS in 1h 21m 13s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/9/check/legacy-grenade-dsvm-neutron/122e163/ : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":9},{"id":"bd90636179d6c15f439baeb04ea15f6076e325bd","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 00:05:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/9/check/check-dsvm-tempest-vz7-exe-minimal/9ce0af9 : SUCCESS in 57m 06s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"b98306b7de7b9ebb88072e8e857fc27ddc21f12b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-21 00:09:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/9/check/dsvm-tempest-neutron-network/eb9e816 : SUCCESS in 1h 53m 57s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/9/check/test-vgpu/48c72fc : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ec03989c30407ac8a4eec9c61b31a296edadfe91","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-21 00:21:36.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"ef349a6a3c4f00838102fb5750162c482720905a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 00:36:36.000000000","message":"Patch Set 9:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/9/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1cea91e : SUCCESS in 1h 41m 12s","accounts_in_message":[],"_revision_number":9},{"id":"fd7cced16dd4635dad809ed1689e500035db677c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-21 01:57:51.000000000","message":"Patch Set 9:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/9/check/dsvm-tempest-xen/441243b : FAILURE in 2h 23m 23s","accounts_in_message":[],"_revision_number":9},{"id":"3f23bbab8c7ac3cd273ed732bb20ad8a997f088e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-21 04:00:24.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/9/check-vote/ext-nova-zuul/935908d : FAILURE in 52m 10s","accounts_in_message":[],"_revision_number":9},{"id":"bdb82eb48973a37236d94191d9bc43d9a04ce130","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-21 15:46:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/9/ : SUCCESS in 3h 27m 26s","accounts_in_message":[],"_revision_number":9},{"id":"e9974587cbf63bf8ae34cd249dc0bd856c286c14","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-25 02:47:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5c43be345e71e539eb6a9f388562ed833296220c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-25 02:49:20.000000000","message":"Patch Set 10:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5d70171 : FAILURE in 30s","accounts_in_message":[],"_revision_number":10},{"id":"aba2349391f1fce263647dca3007093fd375db26","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-25 02:52:25.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/10/check/nova-out-of-tree-pvm/d5063d0 : FAILURE in 3m 48s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/10/check/nova-in-tree-pvm/2b7e81a : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":10},{"id":"b713b7267141c2d06c3565dda3bf59ab8f60c807","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-25 02:57:00.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov-Multinode http://52.169.200.208/23/526823/10/check-mlnx-multinode/Nova-ML2-Sriov-Multinode/9c92baf : FAILURE in 3m 03s (non-voting)\n\nTo re-run the job post \u0027recheck multinode-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"12f3bc29e126df14be1a9fc2e4b11db306bf985a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-25 03:04:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/10/check-nova/Nova-ML2-Sriov/eba49cc : FAILURE in 5m 01s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/10/check-nova/Nova-MACVTAP-ML2-Sriov/8a20f55 : FAILURE in 4m 24s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":10},{"id":"0b6447fb568754bf9c7ee46aa0612e36f5443f1e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-25 03:32:30.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/10/check/tempest-dsvm-full-xenial/c5f1216/ : FAILURE in 8m 34s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/10/check/grenade-dsvm-xenial/1e154ba/ : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"069e758e9e8ade899c29c96eb69f3469b682b74c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-25 03:34:43.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11434 : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":10},{"id":"5e0f096ed4a2cda2e9fde3394da9558d7b80efd5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-25 03:39:24.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/10/check/check-dsvm-tempest-vz7-exe-minimal/479bddf : SUCCESS in 49m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"cd06f22d955bc88427d6675c6d006214f4a3e414","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-25 04:47:07.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/10/check/dsvm-tempest-neutron-network/295c9d3 : SUCCESS in 1h 58m 03s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/10/check/test-vgpu/da71a55 : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ca8359b3a9166410c8b167eedf3ff41f1e43469c","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-25 04:57:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/10/check/dsvm-tempest-xen/a85a04d : SUCCESS in 2h 07m 30s","accounts_in_message":[],"_revision_number":10},{"id":"bd83402a391e015b71aa5b5f370945426d6bbcb3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-25 05:31:12.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/10/ : FAILURE in 2h 42m 32s","accounts_in_message":[],"_revision_number":10},{"id":"183e942e8a368253cc52b3d3f1b779dab6c4f125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-25 06:02:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/10/check/openstack-tox-pep8/10645be/ : SUCCESS in 8m 06s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/10/check/openstack-tox-py27/50de03a/ : SUCCESS in 11m 24s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/10/check/openstack-tox-py35/d09eb22/ : SUCCESS in 10m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/10/check/build-openstack-sphinx-docs/ac1331c/html/ : SUCCESS in 5m 31s\n- tempest-full http://logs.openstack.org/23/526823/10/check/tempest-full/6a0fb98/ : SUCCESS in 1h 22m 04s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/10/check/legacy-grenade-dsvm-neutron/c1d28e2/ : SUCCESS in 42m 00s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-py35/35e1fac/ : SUCCESS in 1h 20m 12s\n- openstack-tox-functional http://logs.openstack.org/23/526823/10/check/openstack-tox-functional/095b614/ : SUCCESS in 20m 34s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/10/check/openstack-tox-functional-py35/1e464c7/ : SUCCESS in 19m 04s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/10/check/legacy-grenade-dsvm-neutron-multinode/4a53f66/ : SUCCESS in 52m 00s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/10/check/legacy-grenade-dsvm-neutron-multinode-live-migration/6e9da57/ : SUCCESS in 50m 15s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-cells/27dc5a0/ : SUCCESS in 35m 07s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/0faf5e5/ : SUCCESS in 58m 39s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-neutron-multinode-full/9ce602c/ : TIMED_OUT in 3h 07m 27s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-neutron-nova-next-full/614fc0b/ : SUCCESS in 1h 09m 51s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/10/check/legacy-tempest-dsvm-multinode-live-migration/c0f7cfe/ : SUCCESS in 45m 58s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/10/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f8ac70f/ : SUCCESS in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"76f27bb77d0a92084915395a332c988de93f24ea","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-25 06:02:38.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"0c7ac2b1e32157a5497ea3552e364ff30172c8d1","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2017-12-27 16:38:57.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"a826ed3cf1de369290ebd4a82223b7aa0a321120","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2017-12-28 01:30:04.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"14b17c11b14ece86c961602c4b941d230dc1ff7a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-28 02:01:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/11/check-nova/Nova-ML2-Sriov/e224764 : FAILURE in 3m 00s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/11/check-nova/Nova-MACVTAP-ML2-Sriov/082a2d6 : FAILURE in 4m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"2022befe209798070e5c08a6d5b409880c9bd3a3","author":{"_account_id":27552,"name":"guozj","email":"guozj@awcloud.com","username":"guozj"},"date":"2017-12-28 02:11:10.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"0f844cfb61d8ff6d52fe112481b80d3bd03ef8d0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-28 02:18:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/11/check/nova-out-of-tree-pvm/429120c : SUCCESS in 48m 05s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/11/check/nova-in-tree-pvm/df32c07 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":11},{"id":"4dd7a24510c95deb764404e3eb53e201179716b5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-28 02:22:32.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/11/check/check-dsvm-tempest-vz7-exe-minimal/0f866f0 : SUCCESS in 51m 01s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"d7e28af1c18a208a9ef2af781f7b14f5e655c162","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-28 02:32:31.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11517 : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":11},{"id":"01ae7260a71a8d9d8a37cb431a69859fd9a50c9c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-28 02:49:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/11/check/tempest-dsvm-full-xenial/b5c3f5d/ : SUCCESS in 1h 02m 36s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/11/check/grenade-dsvm-xenial/4e4e637/ : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"907be481ae3f792dec90d38bf160968ec931d85e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-28 02:59:59.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/0cedf5a : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":11},{"id":"de00d91bf4d9669ddfbe61ab824cff99de0e73b3","author":{"_account_id":27361,"name":"Yandy Sun","email":"sunyd@inspur.com","username":"sunyandi"},"date":"2017-12-28 03:10:09.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"226b7aa2426cbd79ce14d6ba16d45b3fc30be144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 03:11:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/11/check/openstack-tox-pep8/5dab165/ : SUCCESS in 8m 28s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/11/check/openstack-tox-py27/1fcfeb1/ : SUCCESS in 15m 38s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/11/check/openstack-tox-py35/5ba152e/ : SUCCESS in 14m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/11/check/build-openstack-sphinx-docs/9b29e5b/html/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/23/526823/11/check/tempest-full/afb6942/ : SUCCESS in 1h 26m 34s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/11/check/legacy-grenade-dsvm-neutron/5b90790/ : SUCCESS in 46m 04s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-py35/ad58dcd/ : SUCCESS in 1h 13m 00s\n- openstack-tox-functional http://logs.openstack.org/23/526823/11/check/openstack-tox-functional/f6cfa12/ : SUCCESS in 25m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/11/check/openstack-tox-functional-py35/d490ed3/ : SUCCESS in 20m 35s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/11/check/legacy-grenade-dsvm-neutron-multinode/4fcc0a9/ : SUCCESS in 55m 33s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/11/check/legacy-grenade-dsvm-neutron-multinode-live-migration/1bd7a66/ : SUCCESS in 59m 29s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-cells/9ef1980/ : SUCCESS in 36m 21s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/86082e7/ : SUCCESS in 1h 05m 58s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-neutron-multinode-full/373f56b/ : SUCCESS in 1h 37m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-neutron-nova-next-full/4e597da/ : SUCCESS in 1h 19m 16s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/11/check/legacy-tempest-dsvm-multinode-live-migration/6f88ce9/ : SUCCESS in 45m 22s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/11/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b3ee334/ : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d7d266a86f8d2ed454475f9798e4369a1bd47a90","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-28 03:35:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/11/check/dsvm-tempest-neutron-network/044bf87 : SUCCESS in 2h 00m 55s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/11/check/test-vgpu/03e23f1 : SUCCESS in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a3e2591985096abfbc388977aa5aef1dc1c6add7","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-28 03:47:04.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/11/check/dsvm-tempest-xen/1f55c92 : FAILURE in 2h 08m 58s","accounts_in_message":[],"_revision_number":11},{"id":"40c84a456d5f6f252366de70d63cbae42c622d84","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-28 04:18:46.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/11/check-vote/ext-nova-zuul/3d75570 : FAILURE in 51m 53s","accounts_in_message":[],"_revision_number":11},{"id":"4fefed779438a9ba4b4d11fd558e73e97b4fcba1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-28 08:01:30.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/11/ : FAILURE in 6h 13m 48s","accounts_in_message":[],"_revision_number":11},{"id":"3a4abae38cf413a3805e0f942f576f5e2b8b1b49","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2017-12-28 11:24:19.000000000","message":"Patch Set 11: Code-Review+1\n\nLGTM, all request ID of post method have been added.","accounts_in_message":[],"_revision_number":11},{"id":"e806b6705a82d02b13f1de90b56525c75e6bc970","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-02 12:55:03.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"68766dbc1e910480aeadf0ee05adca5e8f556c22","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-02 13:59:21.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/12/check/check-dsvm-tempest-vz7-exe-minimal/471cee6 : SUCCESS in 57m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"e07c70aee2d01ef7a79dd8ae3c7c99bbfa94d6bd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-02 14:02:08.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/tempest-dsvm-full-xenial/4af9131/ : SUCCESS in 1h 05m 32s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/grenade-dsvm-xenial/706f1fe/ : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"35a729c5074459c283e07e5e2a7f19914e4ea34e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-02 14:04:40.000000000","message":"Patch Set 12:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/12/check/nova-out-of-tree-pvm/7e81be1 : SUCCESS in 1h 08m 51s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/12/check/nova-in-tree-pvm/447cd26 : FAILURE in 35m 11s","accounts_in_message":[],"_revision_number":12},{"id":"27a07ca3470e87980493d76029c06292fa94c3f7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-02 14:04:57.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11601 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":12},{"id":"bbd9b13a25ef36d479ff379725f4dbb16f6d4b99","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-01-02 14:51:06.000000000","message":"Patch Set 12:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/a09ee32 : SUCCESS in 1h 53m 51s","accounts_in_message":[],"_revision_number":12},{"id":"27aaf9edfb4540c962819545072bdb515db65da8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-02 15:04:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/12/check/openstack-tox-pep8/1d1df2d/ : SUCCESS in 8m 52s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/12/check/openstack-tox-py27/530d480/ : SUCCESS in 12m 46s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-py35/17dc033/ : SUCCESS in 11m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/12/check/build-openstack-sphinx-docs/929c887/html/ : SUCCESS in 4m 37s\n- tempest-full http://logs.openstack.org/23/526823/12/check/tempest-full/5cdfc6e/ : SUCCESS in 1h 21m 53s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron/e9920e0/ : FAILURE in 1h 25m 05s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-py35/a392b73/ : SUCCESS in 1h 07m 13s\n- openstack-tox-functional http://logs.openstack.org/23/526823/12/check/openstack-tox-functional/47d30ee/ : SUCCESS in 21m 09s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-functional-py35/d1455b9/ : SUCCESS in 22m 00s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode/d8561f6/ : SUCCESS in 54m 47s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode-live-migration/066c1eb/ : SUCCESS in 58m 33s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-cells/6dc41ec/ : SUCCESS in 41m 21s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/8a7993f/ : SUCCESS in 1h 06m 05s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-multinode-full/edc9ded/ : SUCCESS in 1h 28m 04s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-nova-next-full/b62061d/ : SUCCESS in 1h 13m 32s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-multinode-live-migration/c983678/ : SUCCESS in 36m 13s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/12/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/70fa66a/ : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b1abcb4a991ac499a37a205c8ab87ef952bfc853","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2018-01-02 15:09:37.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/23/526823/12/check/dsvm-tempest-xen/cb6e9a0 : SUCCESS in 2h 06m 48s","accounts_in_message":[],"_revision_number":12},{"id":"11f131fe8b859076110425ac653b8502a265142e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-02 15:23:46.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1e2cd3789f8faaa17ebab28051b8fed431f641c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-02 16:21:10.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/tempest-dsvm-full-xenial/6044189/ : SUCCESS in 56m 26s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/grenade-dsvm-xenial/3aa70fe/ : SUCCESS in 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e5c14b0b620ffeb3bab9914e1213afca8dd75ce9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-02 16:33:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/12/ : SUCCESS in 3h 13m 29s","accounts_in_message":[],"_revision_number":12},{"id":"5102db6c9be512ed1afb4f877e608b437a5c48b5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-02 16:38:20.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11623 : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":12},{"id":"5dba29d5704846c5ddc57736155da7bd22cb7ff9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-02 16:45:27.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/12/check/nova-out-of-tree-pvm/1eebfac : SUCCESS in 51m 21s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/12/check/nova-in-tree-pvm/7b66449 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":12},{"id":"c54c7c84e342b80ca864ea60cda711fb8e524ba3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-02 17:04:21.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/12/check-vote/ext-nova-zuul/981e558 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":12},{"id":"af51432d40794b4b52a2cbf67354d5efb08f9b3d","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-02 17:13:13.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/12/check/check-dsvm-tempest-vz7-exe-minimal/a98349c : SUCCESS in 51m 37s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"b6e708df128e5a11f5e7a5f9d138a3490016af67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-02 17:13:56.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/12/check/openstack-tox-pep8/41783d4/ : SUCCESS in 15m 44s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/12/check/openstack-tox-py27/67446c5/ : SUCCESS in 11m 04s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-py35/2948e01/ : SUCCESS in 12m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/12/check/build-openstack-sphinx-docs/2191b5a/html/ : SUCCESS in 5m 44s\n- tempest-full http://logs.openstack.org/23/526823/12/check/tempest-full/4706237/ : SUCCESS in 1h 21m 36s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron/a1d34ab/ : SUCCESS in 44m 05s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-py35/f380928/ : SUCCESS in 1h 17m 32s\n- openstack-tox-functional http://logs.openstack.org/23/526823/12/check/openstack-tox-functional/23e3f34/ : SUCCESS in 24m 37s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-functional-py35/4e8f692/ : FAILURE in 26m 51s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode/dfc9f77/ : SUCCESS in 52m 23s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode-live-migration/001bcd3/ : SUCCESS in 57m 05s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-cells/f471f3d/ : SUCCESS in 40m 28s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6a45e08/ : SUCCESS in 1h 06m 56s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-multinode-full/31fb0e0/ : SUCCESS in 1h 35m 44s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-nova-next-full/a91d1a2/ : SUCCESS in 1h 09m 13s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-multinode-live-migration/628d5fc/ : SUCCESS in 36m 17s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/12/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c94412f/ : SUCCESS in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"86a8055dbca5de570c25a0054dc4938bef3f4d3f","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-03 00:38:53.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"b4b49f88c9c9f27bbe28dcf4fb11bd8f47dbb17b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 01:22:39.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11667 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":12},{"id":"fc5b5879fc5b4f571d23cb5be242b396c9c18211","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-03 01:27:19.000000000","message":"Patch Set 12:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/12/check/nova-out-of-tree-pvm/25ee191 : FAILURE in 47m 44s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/12/check/nova-in-tree-pvm/8b7edd8 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":12},{"id":"f799260503a4fb3433f589305c57986f37b6dc29","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-03 01:31:11.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/12/check/check-dsvm-tempest-vz7-exe-minimal/d39b565 : SUCCESS in 50m 51s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"fe40ba62bb316d0b95500eb44b27aafd8593486b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-03 01:45:29.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/tempest-dsvm-full-xenial/0845b95/ : SUCCESS in 1h 05m 19s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/grenade-dsvm-xenial/ae2674a/ : SUCCESS in 38m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3d024695370543074182e3dc283054c0ea10d6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-03 03:10:36.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/12/check/openstack-tox-pep8/dfe6f1e/ : SUCCESS in 7m 47s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/12/check/openstack-tox-py27/6bfba96/ : SUCCESS in 12m 18s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-py35/82c2230/ : SUCCESS in 11m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/12/check/build-openstack-sphinx-docs/6b213b5/html/ : SUCCESS in 5m 18s\n- tempest-full http://logs.openstack.org/23/526823/12/check/tempest-full/0b2898c/ : SUCCESS in 1h 14m 52s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron/07519f5/ : SUCCESS in 46m 29s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-py35/7e5567e/ : FAILURE in 58m 15s\n- openstack-tox-functional http://logs.openstack.org/23/526823/12/check/openstack-tox-functional/c406c3a/ : SUCCESS in 24m 44s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-functional-py35/a9055af/ : SUCCESS in 19m 48s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode/185127b/ : SUCCESS in 50m 24s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode-live-migration/53fdc10/ : SUCCESS in 56m 06s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-cells/5ae579b/ : FAILURE in 38m 12s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/f38a9d2/ : SUCCESS in 1h 01m 49s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-multinode-full/48fa9de/ : SUCCESS in 1h 30m 06s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-nova-next-full/d7c878d/ : SUCCESS in 1h 16m 27s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-multinode-live-migration/1c20755/ : SUCCESS in 58m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/12/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3cc840f/ : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bc10ad19378c8a5b49fd9f375c56bc498bbf9cc9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-03 04:16:46.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/12/check-vote/ext-nova-zuul/9978737 : FAILURE in 1h 10m 37s","accounts_in_message":[],"_revision_number":12},{"id":"a521573b4f12d094459e4e6303057d93ae80cba7","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-01-03 06:44:56.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/12/check/dsvm-tempest-neutron-network/783377c : SUCCESS in 2h 01m 49s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/12/check/test-vgpu/e7719de : SUCCESS in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d91f0c928fdb8f7825c5182bccf6b924d6697de6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-03 10:07:54.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"7fb94721fe6452c47e299c841f2ce3544e4e996e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-03 10:56:50.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/12/check-vote/ext-nova-zuul/09562a8 : FAILURE in 47m 55s","accounts_in_message":[],"_revision_number":12},{"id":"1e7e633ec45e7223a602c4903b70d779f9f00370","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-03 10:59:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/12/check/check-dsvm-tempest-vz7-exe-minimal/c042fd3 : SUCCESS in 49m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"3e2f796a592d0c7545e0f6fa4e99c32a14f20f51","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-03 11:03:50.000000000","message":"Patch Set 12:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/12/check/nova-out-of-tree-pvm/ec34516 : SUCCESS in 51m 25s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/12/check/nova-in-tree-pvm/efd6645 : FAILURE in 41m 21s","accounts_in_message":[],"_revision_number":12},{"id":"251df80b16c86318ee1e60c0816ae6309802a66c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-03 11:07:10.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/tempest-dsvm-full-xenial/7d0faab/ : SUCCESS in 58m 06s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/12/check/grenade-dsvm-xenial/7de9f51/ : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"eeb9da4c5b83808b875c69c2e54d5341acdd54e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-03 11:35:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/12/check/openstack-tox-pep8/30e709c/ : SUCCESS in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/12/check/openstack-tox-py27/c3a6688/ : SUCCESS in 10m 49s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-py35/0625150/ : SUCCESS in 13m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/12/check/build-openstack-sphinx-docs/bb11611/html/ : SUCCESS in 5m 49s\n- tempest-full http://logs.openstack.org/23/526823/12/check/tempest-full/eb7404c/ : SUCCESS in 1h 24m 48s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron/0fb2d0d/ : SUCCESS in 46m 46s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-py35/c280909/ : SUCCESS in 1h 06m 53s\n- openstack-tox-functional http://logs.openstack.org/23/526823/12/check/openstack-tox-functional/7725e29/ : SUCCESS in 26m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/12/check/openstack-tox-functional-py35/2da252c/ : SUCCESS in 26m 16s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode/dfd5781/ : SUCCESS in 58m 16s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-grenade-dsvm-neutron-multinode-live-migration/287ffe6/ : SUCCESS in 58m 24s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-cells/42eedc3/ : SUCCESS in 35m 36s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/7d85005/ : SUCCESS in 1h 14m 52s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-multinode-full/f98e0c3/ : SUCCESS in 1h 23m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-neutron-nova-next-full/409e55e/ : SUCCESS in 1h 12m 55s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/12/check/legacy-tempest-dsvm-multinode-live-migration/85179a1/ : SUCCESS in 38m 33s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/12/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2195c56/ : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b9695ca424558d19c42cc17a3d230de7df83c819","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 12:16:42.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11686 : FAILURE in 2h 02m 57s","accounts_in_message":[],"_revision_number":12},{"id":"cec1cc2ba60e24bdc23c583cdf3906f85ab5b23e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-03 12:29:40.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/12/check-vote/ext-nova-zuul/784bba7 : FAILURE in 53m 49s","accounts_in_message":[],"_revision_number":12},{"id":"fd9bc2258966ee5114d577e2d448185e9500ff0c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-01-03 12:36:16.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"7c4a8b839a678d4581139ed4b13da42a9309e46b","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 12:41:40.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"9425ca59278c6ebbcf129db87835980b2a509276","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 14:12:16.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11695 : FAILURE in 1h 29m 33s","accounts_in_message":[],"_revision_number":12},{"id":"775d11f456046ec5deb857250c006c7d7e79cd07","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 14:12:37.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"de992367e4af24f7e1125b5463a40105929da4c8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 15:47:55.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11706 : FAILURE in 1h 33m 49s","accounts_in_message":[],"_revision_number":12},{"id":"b475fc612eb6534cf6c35bccaff103fda60c560e","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 16:14:36.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"c70b19d0911b9a608aea4e256ebe6c4440c6f433","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 17:07:16.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11723 : FAILURE in 6m 42s","accounts_in_message":[],"_revision_number":12},{"id":"df407758be12e0995ad208f54e37605468adec66","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 17:15:37.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"edf567eed7d3baf1cfac6937500d525cf541aef3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 19:27:33.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11730 : FAILURE in 1h 32m 59s","accounts_in_message":[],"_revision_number":12},{"id":"fdcfe26ab90cf7c99bf1cd1fdfa187e69055f207","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 19:48:26.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"4e825d2702be04025a64dac05004cdc6a7d453ba","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-03 19:56:53.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/12/ : SUCCESS in 2h 53m 02s","accounts_in_message":[],"_revision_number":12},{"id":"da7345057ae8a2c86e6c1bb99509dc3a929df174","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-03 23:44:31.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11750 : FAILURE in 1h 31m 30s","accounts_in_message":[],"_revision_number":12},{"id":"fbb01ef39c53a667b9cca90a792b27f90ee1be77","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-03 23:53:36.000000000","message":"Patch Set 12:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":12},{"id":"1f8b407338c71d0954ec7de11c31f7bbd1ab9160","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-11 06:09:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4f5b8970556d08703b7063574646ae53f173558a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-11 07:06:30.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/13/check/nova-out-of-tree-pvm/9c2c590 : SUCCESS in 55m 59s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/13/check/nova-in-tree-pvm/3bc778c : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":13},{"id":"e355b4af81a822272b38cfa4ffd007bf0e91bac2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-11 07:08:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/13/check/tempest-dsvm-full-xenial/3ea4895/ : SUCCESS in 57m 27s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/13/check/grenade-dsvm-xenial/b7c365e/ : SUCCESS in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6a3bf82f8f1cd4e855747f739ee285af3fd65928","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-11 07:09:03.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-12055 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":13},{"id":"60b90b906e5ad089f8b7f73d4bf2474164103660","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-01-11 07:46:05.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2ba68dc : SUCCESS in 1h 35m 27s","accounts_in_message":[],"_revision_number":13},{"id":"9bfb7621a45214d9fd5dcf8ba1824204ba709df9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-01-11 07:55:00.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/13/check-nova/Nova-ML2-Sriov/9e2f65e : SUCCESS in 1h 00m 14s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/13/check-nova/Nova-MACVTAP-ML2-Sriov/fa86e1f : SUCCESS in 48m 24s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"225f6fbe921f159d92e10d0010a3b6ab391270c9","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-01-11 08:08:02.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/13/check/dsvm-tempest-neutron-network/46a317f : SUCCESS in 1h 57m 07s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/13/check/test-vgpu/c5b682f : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"437ac9092ce1014b72bee9127275e0131f5b7709","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-11 08:18:06.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/13/check/check-dsvm-tempest-vz7-exe-minimal/c281c2e : SUCCESS in 52m 47s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"f5017b81d3e88ab64ebeb8a49341a9d3517f6a20","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-11 09:44:11.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/13/ : SUCCESS in 2h 16m 38s","accounts_in_message":[],"_revision_number":13},{"id":"caa56006a78ba8d62827ea4ad11801f1a3e2a8cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 11:27:59.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/13/check/openstack-tox-pep8/7668409/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/13/check/openstack-tox-py27/9244236/ : SUCCESS in 12m 19s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/13/check/openstack-tox-py35/11974b3/ : SUCCESS in 14m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/13/check/build-openstack-sphinx-docs/55ae34d/html/ : SUCCESS in 5m 51s\n- tempest-full http://logs.openstack.org/23/526823/13/check/tempest-full/2a83435/ : SUCCESS in 1h 39m 50s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/23/526823/13/check/legacy-grenade-dsvm-neutron/febad3a/ : SUCCESS in 44m 54s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-py35/0742276/ : SUCCESS in 1h 13m 17s\n- openstack-tox-functional http://logs.openstack.org/23/526823/13/check/openstack-tox-functional/3aaf9c8/ : SUCCESS in 24m 22s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/13/check/openstack-tox-functional-py35/80e8bdf/ : SUCCESS in 24m 05s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/23/526823/13/check/legacy-grenade-dsvm-neutron-multinode/f1628b0/ : SUCCESS in 1h 04m 23s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/13/check/legacy-grenade-dsvm-neutron-multinode-live-migration/91dee5e/ : SUCCESS in 50m 15s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-cells/0112ba1/ : SUCCESS in 39m 35s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/ccf4451/ : SUCCESS in 1h 02m 34s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-neutron-multinode-full/cf39f05/ : SUCCESS in 1h 35m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-neutron-nova-next-full/48aba24/ : SUCCESS in 1h 31m 46s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/13/check/legacy-tempest-dsvm-multinode-live-migration/b2af19b/ : SUCCESS in 37m 00s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/13/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/bf52dce/ : SUCCESS in 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1ffeeea48a9e5b5a3c3cf4f9f3c106076899c651","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-01-11 12:50:50.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"377d4394c4a77a8474a0a504e0ae02b46deae4f5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-11 14:25:27.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/13/check-vote/ext-nova-zuul/3ad9243 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":13},{"id":"fc4a68f085bb6f0c650b92aabe110fad67a8ce6b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-01-12 02:50:45.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"bebc013029bd7176841dfcb6d6bdb647df3e71b4","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-18 04:36:38.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"eb45afefdd22b4c00729c37424ce4694642fc374","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-18 05:29:57.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/14/check/nova-out-of-tree-pvm/3be9951 : SUCCESS in 47m 28s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/14/check/nova-in-tree-pvm/acc4129 : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":14},{"id":"34405e18986912361f3aa138e4c249b9e3aa9b66","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-18 05:34:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/14/check/tempest-dsvm-full-xenial/a19a6be/ : SUCCESS in 56m 27s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/14/check/grenade-dsvm-xenial/b30079b/ : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f66c7d4b08c34ddafa45b8a27a74fae0b24ae57e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-18 05:43:35.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-12685 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":14},{"id":"a7a96578683f8da4de662a329a3e4ddec014012c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-01-18 06:13:55.000000000","message":"Patch Set 14:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/14/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c8dd235 : SUCCESS in 1h 36m 37s","accounts_in_message":[],"_revision_number":14},{"id":"354298aca300738d58a8b4e1e54a5a5132f8874a","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-01-18 06:42:32.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/14/check/dsvm-tempest-neutron-network/1d02d7e : SUCCESS in 2h 05m 02s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/14/check/test-vgpu/3b532e3 : SUCCESS in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d2209628a32912a71b0f609b08e4eaea52354e33","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-18 07:14:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/14/ : SUCCESS in 2h 30m 42s","accounts_in_message":[],"_revision_number":14},{"id":"c23ff77492fffc00835b3a794df8e1a4820009cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-18 07:48:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/14/check/openstack-tox-pep8/611693e/ : SUCCESS in 11m 30s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/14/check/openstack-tox-py27/20e9307/ : SUCCESS in 12m 17s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/14/check/openstack-tox-py35/95cc611/ : SUCCESS in 16m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/14/check/build-openstack-sphinx-docs/35a9f0c/html/ : SUCCESS in 6m 39s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-neutron-full/efaa6fb/ : SUCCESS in 1h 23m 10s\n- tempest-full http://logs.openstack.org/23/526823/14/check/tempest-full/6ad3ff8/ : SUCCESS in 1h 31m 18s\n- neutron-grenade http://logs.openstack.org/23/526823/14/check/neutron-grenade/38a1680/ : SUCCESS in 44m 56s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-py35/13270e7/ : SUCCESS in 1h 26m 51s\n- openstack-tox-functional http://logs.openstack.org/23/526823/14/check/openstack-tox-functional/85a2eea/ : SUCCESS in 29m 03s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/14/check/openstack-tox-functional-py35/b558951/ : SUCCESS in 24m 25s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/14/check/neutron-grenade-multinode/71837a3/ : SUCCESS in 1h 06m 50s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/14/check/legacy-grenade-dsvm-neutron-multinode-live-migration/0ed8c07/ : SUCCESS in 58m 25s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-cells/ba11995/ : SUCCESS in 56m 50s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/408e138/ : SUCCESS in 2h 57m 09s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/14/check/neutron-tempest-multinode-full/0f5c4ce/ : SUCCESS in 1h 43m 29s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-neutron-nova-next-full/c94995c/ : SUCCESS in 1h 21m 27s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/14/check/legacy-tempest-dsvm-multinode-live-migration/af2c5d0/ : SUCCESS in 43m 11s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/14/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1e4b5e9/ : SUCCESS in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a2a67d0742f967be9e4a7276c02ba148dd71460f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-18 08:59:10.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/14/check-vote/ext-nova-zuul/2de4eb5 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":14},{"id":"19360555ae05abef39fbfc1976fb2931a2ffda92","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-18 18:28:09.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/14/check/check-dsvm-tempest-vz7-exe-minimal/d0979d3 : SUCCESS in 53m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"a2bce22a044df2e2471555c1b88739c3fc6d8831","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-01-22 03:17:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"66281979a6547bd4077f82b5833f456c884c6f8c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-22 03:19:02.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-None : NOT_REGISTERED","accounts_in_message":[],"_revision_number":15},{"id":"959f838ce6bb678ce38ceb568def3cbe5e0abb48","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-22 04:07:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/15/check/nova-out-of-tree-pvm/f61a2aa : SUCCESS in 46m 46s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/15/check/nova-in-tree-pvm/afda906 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":15},{"id":"612a51d29c5796d2a66b6e8d50105db66434b1a2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-01-22 04:14:12.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/15/check/tempest-dsvm-full-xenial/c61a77d/ : SUCCESS in 55m 42s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/15/check/grenade-dsvm-xenial/eb7d106/ : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e8cf1b8a04fa82f44931d2f8fdf79680c19f6577","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-01-22 04:44:32.000000000","message":"Patch Set 15:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e7d57ad : SUCCESS in 1h 26m 28s","accounts_in_message":[],"_revision_number":15},{"id":"f56360ff3335a4c020999f31b57d0ecb20a09095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 04:55:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/15/check/openstack-tox-pep8/0f3982a/ : SUCCESS in 17m 47s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/15/check/openstack-tox-py27/9def63f/ : SUCCESS in 18m 02s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/15/check/openstack-tox-py35/10b1360/ : SUCCESS in 15m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/15/check/build-openstack-sphinx-docs/2b258d3/html/ : SUCCESS in 7m 33s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-neutron-full/04a99ea/ : SUCCESS in 1h 34m 09s\n- tempest-full http://logs.openstack.org/23/526823/15/check/tempest-full/17086d5/ : SUCCESS in 1h 20m 06s\n- neutron-grenade http://logs.openstack.org/23/526823/15/check/neutron-grenade/dec5b94/ : SUCCESS in 48m 49s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-py35/5fb1975/ : SUCCESS in 1h 14m 25s\n- openstack-tox-functional http://logs.openstack.org/23/526823/15/check/openstack-tox-functional/ca46957/ : SUCCESS in 27m 57s\n- openstack-tox-functional-py35 http://logs.openstack.org/23/526823/15/check/openstack-tox-functional-py35/7795043/ : SUCCESS in 22m 38s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/15/check/neutron-grenade-multinode/1a574d6/ : SUCCESS in 54m 53s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/15/check/legacy-grenade-dsvm-neutron-multinode-live-migration/68e92c7/ : SUCCESS in 1h 04m 36s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-cells/81f703e/ : SUCCESS in 43m 55s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/72dc708/ : SUCCESS in 1h 10m 37s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/15/check/neutron-tempest-multinode-full/9a7fcae/ : SUCCESS in 1h 33m 12s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-neutron-nova-next-full/d5cfd83/ : SUCCESS in 1h 14m 38s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/15/check/legacy-tempest-dsvm-multinode-live-migration/16c8c82/ : SUCCESS in 40m 26s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/15/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6a651b9/ : SUCCESS in 45m 57s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c3ee45c9580632987a49d9cd4e474b763a9b451a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-01-22 05:08:39.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/15/check/check-dsvm-tempest-vz7-exe-minimal/0034d64 : SUCCESS in 53m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"32763d87a971a70a44bb4af6646c357f557715ed","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-01-22 05:38:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/15/check/dsvm-tempest-neutron-network/ba093a4 : SUCCESS in 2h 01m 54s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/15/check/test-vgpu/76d194f : SUCCESS in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"bfa9b4401878d8ef089e9cc1759c7f581149abc7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-01-22 06:23:24.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"a121c4abeb6efc3ee8f1887c5f14c55f5d7906dc","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2018-01-22 06:24:30.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"9288a2e5c21e4169367aabc8111d6a989a4ff3a6","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-22 09:01:32.000000000","message":"Patch Set 15:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":15},{"id":"f9d177eafa50b9b0b4da05036d6d2f6bae530a23","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-22 09:45:17.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/15/ : FAILURE in 6h 27m 14s","accounts_in_message":[],"_revision_number":15},{"id":"f86e0664fac9a55a7d9747b9c6304e0f2c7ffc31","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-22 11:22:30.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-12756 : FAILURE in 2h 04m 41s","accounts_in_message":[],"_revision_number":15},{"id":"f82bf748b18122d82d1c3e399fb34cde05db6f14","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-22 11:29:01.000000000","message":"Patch Set 15:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":15},{"id":"71d80e72e459db38b88bb75efbfb379eb79c2bc6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-01-22 14:48:08.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"bb234c4f08c88d4a9055b52848d793b1aa66fcea","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2018-01-23 07:12:34.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"e6845369ddaf3e7465d1a91abe495f9d58ba3342","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-01-23 08:01:02.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/23/526823/15/check-vote/ext-nova-zuul/d87a429 : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":15},{"id":"2dfc68aed11b8bba70503931790af92db0bdbb8a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-23 14:11:11.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-12778 : FAILURE in 3h 59m 52s","accounts_in_message":[],"_revision_number":15},{"id":"ed64d8e9bb487b59eff5e053c9d5c2457a7b2f45","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-23 17:41:24.000000000","message":"Patch Set 15:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":15},{"id":"7105f291e7d389e467ffeeea44c9267a5230bacf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-26 13:31:44.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-12857 : FAILURE in 4h 21m 30s","accounts_in_message":[],"_revision_number":15},{"id":"160fa93821d921265b983295b2bba265cf73c3e1","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-26 21:48:13.000000000","message":"Patch Set 15:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":15},{"id":"0d3c14a7153025dc3c610ff667d306658d3a7d58","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-27 12:16:15.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13024 : FAILURE in 3h 38m 17s","accounts_in_message":[],"_revision_number":15},{"id":"4b7145b0532909f4359a642c9d6bec13cc6512a9","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-01-27 12:54:48.000000000","message":"Patch Set 15:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":15},{"id":"4f0aee13f0ed2a28eb95d74466e5d763adbd44d8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-27 18:00:49.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13063 : SUCCESS in 2h 13m 39s","accounts_in_message":[],"_revision_number":15},{"id":"4b88f6da4c96296619f0b6e02504c9c669f5728a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-02-01 22:19:57.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"d979c116787690ee37a6c8508e367f01008dc466","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-01 23:18:40.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/16/check/nova-out-of-tree-pvm/70ea3a8 : SUCCESS in 58m 10s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/16/check/nova-in-tree-pvm/d109e57 : SUCCESS in 46m 06s","accounts_in_message":[],"_revision_number":16},{"id":"0f902412abed112f97f662612b9f158f92e950b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-01 23:19:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/16/check/tempest-dsvm-full-xenial/3974771/ : SUCCESS in 58m 34s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/16/check/grenade-dsvm-xenial/7109b79/ : SUCCESS in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f5d8be086b73c4bb31b10c6be9e47dc520755389","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-01 23:31:32.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/16/check/check-dsvm-tempest-vz7-exe-minimal/540b671 : SUCCESS in 1h 10m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"106fbce3336033b4a9a14bf8a13995edc6c9e3e7","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-01 23:58:24.000000000","message":"Patch Set 16:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/16/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/cbe3c1b : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":16},{"id":"37cb708d2a1ce50708e3596248653cf6bb8dbbce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-02 00:49:05.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/16/check/openstack-tox-pep8/978e009/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/16/check/openstack-tox-py27/f977006/ : SUCCESS in 13m 12s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/16/check/openstack-tox-py35/6691a5f/ : SUCCESS in 16m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/16/check/build-openstack-sphinx-docs/cdbb69f/html/ : SUCCESS in 5m 48s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-neutron-full/824576f/ : SUCCESS in 1h 35m 39s\n- tempest-full http://logs.openstack.org/23/526823/16/check/tempest-full/51e3d40/ : SUCCESS in 1h 15m 01s\n- neutron-grenade http://logs.openstack.org/23/526823/16/check/neutron-grenade/af3761b/ : SUCCESS in 50m 02s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-py35/3eb107d/ : SUCCESS in 1h 08m 41s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/16/check/neutron-grenade-multinode/a8d377b/ : SUCCESS in 52m 54s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/16/check/legacy-grenade-dsvm-neutron-multinode-live-migration/372319b/ : SUCCESS in 1h 02m 43s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-cells/2eae84a/ : SUCCESS in 43m 22s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/313b3fe/ : SUCCESS in 1h 04m 19s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/16/check/neutron-tempest-multinode-full/6238dd7/ : SUCCESS in 1h 53m 33s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-neutron-nova-next-full/9672c5d/ : FAILURE in 48m 23s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/16/check/legacy-tempest-dsvm-multinode-live-migration/f6e3d37/ : SUCCESS in 39m 55s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/16/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b4fc0ae/ : SUCCESS in 36m 43s (non-voting)\n- nova-tox-functional http://logs.openstack.org/23/526823/16/check/nova-tox-functional/12caa96/ : FAILURE in 21m 51s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/16/check/nova-tox-functional-py35/2344aac/ : FAILURE in 26m 13s","accounts_in_message":[],"_revision_number":16},{"id":"3aef4fa2429bf9c7bf20d087d2b7b798cb19eb8d","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-02-02 01:09:29.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/23/526823/16/check-nova/Nova-ML2-Sriov/5393639 : SUCCESS in 1h 08m 32s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/23/526823/16/check-nova/Nova-MACVTAP-ML2-Sriov/a2be3e6 : SUCCESS in 47m 13s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":16},{"id":"47c787097d290a531404e6e866b73acc6d6bd5e2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-02 01:17:01.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/16/check/dsvm-tempest-neutron-network/20ecdcc : FAILURE in 2h 53m 05s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/16/check/test-vgpu/cad1652 : SUCCESS in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"492c4ca6f2c9a101d2c858e74c05a09357af236b","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-02-02 01:37:27.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f79dfe5704656ce8bb26dafe912d6732bec6c2c8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-02 02:37:21.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/17/check/check-dsvm-tempest-vz7-exe-minimal/ac9ef6c : SUCCESS in 51m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"3300a2dec772a6d60a038ebec879c8e7ba626724","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-02 02:38:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/17/check/tempest-dsvm-full-xenial/e71237a/ : SUCCESS in 1h 00m 28s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/17/check/grenade-dsvm-xenial/addf979/ : SUCCESS in 46m 48s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f07b55fc4eaef200de7914fc87301cef86bbddb2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-02 02:41:08.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/17/check/nova-out-of-tree-pvm/0409219 : SUCCESS in 1h 02m 45s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/17/check/nova-in-tree-pvm/bc10f28 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":17},{"id":"0c5c6d51307679f4b757ebe779b852900611cb2b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-02 03:01:31.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13565 : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":17},{"id":"d208d6012ab545d8bc75f6171fd6acb10f2ef2ad","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-02 03:10:53.000000000","message":"Patch Set 17:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/8b0608c : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":17},{"id":"04e84bc30b9b8edc2c31640044b605a64b1a2913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-02 03:13:22.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/17/check/openstack-tox-pep8/8ee77b0/ : SUCCESS in 9m 08s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/17/check/openstack-tox-py27/ce02329/ : SUCCESS in 11m 32s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/17/check/openstack-tox-py35/06271dd/ : SUCCESS in 18m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/17/check/build-openstack-sphinx-docs/e342cdb/html/ : SUCCESS in 6m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-neutron-full/c1c7707/ : SUCCESS in 1h 17m 41s\n- tempest-full http://logs.openstack.org/23/526823/17/check/tempest-full/e8c2a4d/ : SUCCESS in 1h 21m 14s\n- neutron-grenade http://logs.openstack.org/23/526823/17/check/neutron-grenade/715e4da/ : SUCCESS in 1h 05m 53s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-py35/8a9445a/ : SUCCESS in 1h 11m 14s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/17/check/neutron-grenade-multinode/cf43328/ : SUCCESS in 1h 01m 22s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/17/check/legacy-grenade-dsvm-neutron-multinode-live-migration/710c0a3/ : SUCCESS in 1h 01m 12s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-cells/39ae16e/ : SUCCESS in 42m 55s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/8a7e1c2/ : SUCCESS in 1h 27m 30s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/17/check/neutron-tempest-multinode-full/cfa77b6/ : FAILURE in 1h 20m 02s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-neutron-nova-next-full/7d5104f/ : FAILURE in 49m 27s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/17/check/legacy-tempest-dsvm-multinode-live-migration/05f3e7f/ : SUCCESS in 36m 26s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/17/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/aa93670/ : SUCCESS in 55m 27s (non-voting)\n- nova-tox-functional http://logs.openstack.org/23/526823/17/check/nova-tox-functional/f39432a/ : SUCCESS in 21m 54s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/17/check/nova-tox-functional-py35/c31c990/ : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":17},{"id":"458ad55b2e064775e4cf680318d2e3d26e1bdc94","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-02 03:26:55.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/17/check/dsvm-tempest-neutron-network/95216e8 : SUCCESS in 1h 48m 14s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/17/check/test-vgpu/e667200 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c42aec90bf82afdb1154a01f36855134562493de","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-02 07:23:58.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/526823/17/ : FAILURE in 5h 38m 32s","accounts_in_message":[],"_revision_number":17},{"id":"34344ec28b3dcbd16ee2e40f21704eb7388b23a2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-02-02 08:07:15.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"cbe4c4bd69147deb5742750c4235fd29eaa0c348","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-02-04 23:40:53.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"1d1a5ad3f2459d9e4b9ec2c0a1a694413a49c778","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-05 00:40:59.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/18/check/tempest-dsvm-full-xenial/354fb19/ : SUCCESS in 58m 33s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/18/check/grenade-dsvm-xenial/33577c8/ : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0947051f0754f4405eb195d774242e05dbb00af9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-05 00:51:06.000000000","message":"Patch Set 18:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13644 : FAILURE in 1h 09m 45s","accounts_in_message":[],"_revision_number":18},{"id":"1194bd53b5ae433b58d9f07e5c000e7dc2705398","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-05 01:20:04.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/18/check/nova-out-of-tree-pvm/683279f : SUCCESS in 1h 38m 35s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/18/check/nova-in-tree-pvm/830c5cd : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":18},{"id":"a6da11b976b72def36025881d0c9f52816279172","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-02-05 01:26:46.000000000","message":"Patch Set 18:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/526823/18/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/3562946 : SUCCESS in 1h 45m 19s","accounts_in_message":[],"_revision_number":18},{"id":"a9ecc0f32de1e799ab87e4bda0c2b0db3559ba73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-05 01:32:24.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/18/check/openstack-tox-pep8/6196bc4/ : SUCCESS in 9m 37s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/18/check/openstack-tox-py27/2915709/ : SUCCESS in 13m 46s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/18/check/openstack-tox-py35/4ce3cda/ : SUCCESS in 12m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/18/check/build-openstack-sphinx-docs/e7a7628/html/ : SUCCESS in 5m 44s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-neutron-full/12fb8f3/ : SUCCESS in 1h 18m 14s\n- tempest-full http://logs.openstack.org/23/526823/18/check/tempest-full/0ab5f50/ : SUCCESS in 1h 30m 40s\n- neutron-grenade http://logs.openstack.org/23/526823/18/check/neutron-grenade/7727516/ : SUCCESS in 42m 32s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-py35/61412dc/ : SUCCESS in 1h 21m 27s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/18/check/neutron-grenade-multinode/1345dd8/ : SUCCESS in 1h 12m 33s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/18/check/legacy-grenade-dsvm-neutron-multinode-live-migration/e9acc0b/ : SUCCESS in 56m 27s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-cells/fe98e33/ : SUCCESS in 39m 17s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/2f2bc2f/ : SUCCESS in 57m 24s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/18/check/neutron-tempest-multinode-full/1db8d9c/ : SUCCESS in 1h 49m 35s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-neutron-nova-next-full/b2a8302/ : FAILURE in 50m 50s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-multinode-live-migration/0f5e80c/ : SUCCESS in 38m 18s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/18/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9d8401e/ : SUCCESS in 37m 11s (non-voting)\n- nova-multiattach http://logs.openstack.org/23/526823/18/check/nova-multiattach/69592ab/ : SUCCESS in 46m 13s\n- nova-tox-functional http://logs.openstack.org/23/526823/18/check/nova-tox-functional/82fa326/ : SUCCESS in 28m 35s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/18/check/nova-tox-functional-py35/f1f7fe9/ : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":18},{"id":"354c99e7b92bfa0e738f27f1a426d110588ba58f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-05 01:37:01.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/18/check/check-dsvm-tempest-vz7-exe-minimal/d30710f : SUCCESS in 51m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"546a794f1183eb7cca190a8aaea20ea409c1b74f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-05 01:58:16.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/18/ : SUCCESS in 2h 16m 52s","accounts_in_message":[],"_revision_number":18},{"id":"d25c2913a6991b591adb7873a668a1e5c1f25c20","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-02-05 02:17:11.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/18/check/dsvm-tempest-neutron-network/ff97ecf : SUCCESS in 2h 19m 35s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/23/526823/18/check/test-vgpu/3b761e1 : SUCCESS in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e05ac55aa90e6f1da7669f00f2532fea57dd9665","author":{"_account_id":6125,"name":"Huang Shi Lin","email":"shilinh@cn.ibm.com","username":"shilinh"},"date":"2018-02-05 05:41:02.000000000","message":"Patch Set 18:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":18},{"id":"1272682844b8fd4c6f7ed64aa027c260dbf1e764","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-05 07:00:42.000000000","message":"Patch Set 18:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13670 : SUCCESS in 1h 17m 44s","accounts_in_message":[],"_revision_number":18},{"id":"5b61d9faf86fbb963fd63c1f198a55f29400a355","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-05 12:33:56.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)\n\nLGTM!","accounts_in_message":[],"_revision_number":18},{"id":"b250f30fea3d4199c3b9518adbd7c78276aa99e7","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-02-05 16:39:28.000000000","message":"Patch Set 18: Workflow+1 Code-Review+2\n\n++","accounts_in_message":[],"_revision_number":18},{"id":"e1d74178c0f647d0cd00d705f141f25633e8e488","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-05 16:40:18.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"e2b3b29dd4496bedc8535df0daf023bbeefa47de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-05 20:40:24.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/18/gate/openstack-tox-pep8/07ce44d/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/18/gate/openstack-tox-py27/dd4e672/ : SUCCESS in 11m 01s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/18/gate/openstack-tox-py35/f4dde76/ : SUCCESS in 22m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/18/gate/build-openstack-sphinx-docs/acb2add/html/ : SUCCESS in 5m 22s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-neutron-full/58f601b/ : SUCCESS in 1h 17m 10s\n- tempest-full http://logs.openstack.org/23/526823/18/gate/tempest-full/7cb8624/ : TIMED_OUT in 2h 18m 08s\n- neutron-grenade http://logs.openstack.org/23/526823/18/gate/neutron-grenade/95e9bdb/ : SUCCESS in 45m 23s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-py35/d46bece/ : SUCCESS in 1h 08m 39s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-cells/981ad4f/ : SUCCESS in 1h 07m 35s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-multinode-live-migration/c3c76b2/ : SUCCESS in 41m 30s\n- nova-multiattach http://logs.openstack.org/23/526823/18/gate/nova-multiattach/05779bd/ : SUCCESS in 45m 47s\n- nova-tox-functional http://logs.openstack.org/23/526823/18/gate/nova-tox-functional/296b704/ : SUCCESS in 23m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/18/gate/nova-tox-functional-py35/6442885/ : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":18},{"id":"0a29667a352495fe339eac0239982bd0e96d7861","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-02-05 23:57:41.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"bc00fbde9acaf68b85e2008956ce88dabf032ad9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-02-06 00:56:34.000000000","message":"Patch Set 18:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13772 : FAILURE in 46m 49s","accounts_in_message":[],"_revision_number":18},{"id":"d2b13d1dbb2efe333aae9b4fc8ffa59d9cc55f3d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-02-06 01:05:32.000000000","message":"Patch Set 18:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/18/check/tempest-dsvm-full-xenial/b52d19d/ : SUCCESS in 57m 36s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/23/526823/18/check/grenade-dsvm-xenial/023448e/ : SUCCESS in 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"9178bb6d1a6611d7632021b0263b0f3e1c935f2e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-02-06 01:14:10.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/23/526823/18/check/nova-out-of-tree-pvm/5124ca6 : SUCCESS in 55m 05s\n- nova-in-tree-pvm http://184.172.12.213/23/526823/18/check/nova-in-tree-pvm/577b321 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":18},{"id":"9ff5cacb2cb8b588a20fcd944771a2d19ec6483a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 01:53:00.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/18/check/openstack-tox-pep8/7ae82f3/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/18/check/openstack-tox-py27/37287fe/ : SUCCESS in 12m 27s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/18/check/openstack-tox-py35/9fb0f3a/ : SUCCESS in 14m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/18/check/build-openstack-sphinx-docs/026b1bc/html/ : SUCCESS in 6m 39s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-neutron-full/06d51c3/ : SUCCESS in 1h 17m 23s\n- tempest-full http://logs.openstack.org/23/526823/18/check/tempest-full/b70134d/ : SUCCESS in 1h 33m 39s\n- neutron-grenade http://logs.openstack.org/23/526823/18/check/neutron-grenade/a8d6a7a/ : SUCCESS in 48m 54s\n- tempest-full-py3 http://logs.openstack.org/23/526823/18/check/tempest-full-py3/60e4468/ : SUCCESS in 1h 03m 09s\n- neutron-grenade-multinode http://logs.openstack.org/23/526823/18/check/neutron-grenade-multinode/4218e8c/ : SUCCESS in 53m 29s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/23/526823/18/check/legacy-grenade-dsvm-neutron-multinode-live-migration/93f4d39/ : SUCCESS in 54m 40s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-cells/aa1d637/ : SUCCESS in 43m 31s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/2f7f7db/ : SUCCESS in 53m 57s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/23/526823/18/check/neutron-tempest-multinode-full/0b2661e/ : SUCCESS in 1h 48m 25s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-neutron-nova-next-full/a67cce2/ : FAILURE in 1h 03m 58s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/18/check/legacy-tempest-dsvm-multinode-live-migration/561804d/ : SUCCESS in 38m 35s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/23/526823/18/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2a275fc/ : SUCCESS in 44m 00s (non-voting)\n- nova-multiattach http://logs.openstack.org/23/526823/18/check/nova-multiattach/9d026c4/ : SUCCESS in 50m 31s\n- nova-tox-functional http://logs.openstack.org/23/526823/18/check/nova-tox-functional/5e8c36e/ : SUCCESS in 20m 26s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/18/check/nova-tox-functional-py35/9827f2f/ : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":18},{"id":"f628839bdb441c816e7e2d390b47e018417a3e09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 01:53:19.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"bb20bc2512df98e7ee1e77f39ef0e3bb9e068b75","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-02-06 02:23:48.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/23/526823/18/check/check-dsvm-tempest-vz7-exe-minimal/71762e2 : SUCCESS in 54m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":18},{"id":"0e0e2a85e306682e20a1d42bd6985b81d0fca0d2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-06 03:33:03.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/526823/18/ : SUCCESS in 2h 36m 39s","accounts_in_message":[],"_revision_number":18},{"id":"471b5eb82fabb35dc53d1f607c6007d3df840ee7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 04:05:40.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/526823/18/gate/openstack-tox-pep8/295f1b3/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/23/526823/18/gate/openstack-tox-py27/266fac5/ : SUCCESS in 11m 32s\n- openstack-tox-py35 http://logs.openstack.org/23/526823/18/gate/openstack-tox-py35/743634b/ : SUCCESS in 13m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/526823/18/gate/build-openstack-sphinx-docs/4b9039f/html/ : SUCCESS in 4m 53s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-neutron-full/831d3e2/ : SUCCESS in 1h 29m 58s\n- tempest-full http://logs.openstack.org/23/526823/18/gate/tempest-full/ead10f6/ : SUCCESS in 1h 35m 34s\n- neutron-grenade http://logs.openstack.org/23/526823/18/gate/neutron-grenade/45089b6/ : SUCCESS in 51m 13s\n- tempest-full-py3 http://logs.openstack.org/23/526823/18/gate/tempest-full-py3/ff98c52/ : SUCCESS in 1h 15m 08s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-cells/4ed477c/ : SUCCESS in 37m 12s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/23/526823/18/gate/legacy-tempest-dsvm-multinode-live-migration/b9076c1/ : SUCCESS in 43m 13s\n- nova-multiattach http://logs.openstack.org/23/526823/18/gate/nova-multiattach/84c9271/ : SUCCESS in 55m 46s\n- nova-tox-functional http://logs.openstack.org/23/526823/18/gate/nova-tox-functional/22c20da/ : SUCCESS in 19m 34s\n- nova-tox-functional-py35 http://logs.openstack.org/23/526823/18/gate/nova-tox-functional-py35/7f995b1/ : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":18},{"id":"7c39bed56cd9e1056bac36c2f375669bf5e054f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-06 04:05:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"6bd7f0823bd310e21f573febbc67fdff9b384947","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-02-06 04:51:51.000000000","message":"Patch Set 18:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-23-526823-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"6079b722cbfc96dfab884a8350d8ade67e3fb840","revisions":{"a803e9c179c8a941b77658b15721938ea43b396d":{"kind":"REWORK","_number":1,"created":"2017-12-09 06:26:47.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/1"}}},"commit":{"parents":[{"commit":"9a9dc626283f6c6bd0a78ff9f1a9ccafb261d3a5","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9a9dc626283f6c6bd0a78ff9f1a9ccafb261d3a5"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 06:25:42.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a803e9c179c8a941b77658b15721938ea43b396d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a803e9c179c8a941b77658b15721938ea43b396d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a9dc626283f6c6bd0a78ff9f1a9ccafb261d3a5","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"004643ff5118b74a8056a80cf3eb7bd7693a9c0a":{"kind":"REWORK","_number":2,"created":"2017-12-09 08:23:47.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/2"}}},"commit":{"parents":[{"commit":"f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 08:23:07.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/004643ff5118b74a8056a80cf3eb7bd7693a9c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/004643ff5118b74a8056a80cf3eb7bd7693a9c0a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a8830b3542d9a413037d3997121f5bc6a8e10a98":{"kind":"REWORK","_number":3,"created":"2017-12-11 00:50:05.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/3"}}},"commit":{"parents":[{"commit":"f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-11 00:49:31.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8830b3542d9a413037d3997121f5bc6a8e10a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8830b3542d9a413037d3997121f5bc6a8e10a98"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f7a7ad1d5fe0a3af1203f8afbbc29d00d03934a6","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e55612d6233115ca3687d6c47a33bd23881be748":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-12-11 00:50:09.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/4"}}},"commit":{"parents":[{"commit":"d9b9500f9173000e785f836e91795e17cd0d4ae8","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9b9500f9173000e785f836e91795e17cd0d4ae8"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-11 00:50:09.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e55612d6233115ca3687d6c47a33bd23881be748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e55612d6233115ca3687d6c47a33bd23881be748"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9b9500f9173000e785f836e91795e17cd0d4ae8","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"60b23889ba9a1ab5862fa86f70c5703cd16eed22":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2017-12-12 06:10:21.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/5"}}},"commit":{"parents":[{"commit":"d6039d68871f7540360f75b10e2b6d59137f426f","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6039d68871f7540360f75b10e2b6d59137f426f"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-12 06:10:21.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60b23889ba9a1ab5862fa86f70c5703cd16eed22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60b23889ba9a1ab5862fa86f70c5703cd16eed22"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d6039d68871f7540360f75b10e2b6d59137f426f","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d606fab816900f2d55b9927cd084b0fba6a21521":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2017-12-13 02:06:19.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/6"}}},"commit":{"parents":[{"commit":"65a23ba7752218ad16ccc3369ed1b9a26c1d4631","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65a23ba7752218ad16ccc3369ed1b9a26c1d4631"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-13 02:06:19.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d606fab816900f2d55b9927cd084b0fba6a21521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d606fab816900f2d55b9927cd084b0fba6a21521"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"65a23ba7752218ad16ccc3369ed1b9a26c1d4631","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"15c62db3697ed6e460f214acdaf40757fcbb9777":{"kind":"REWORK","_number":7,"created":"2017-12-14 09:17:52.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/7"}}},"commit":{"parents":[{"commit":"ea6a89104a3d3974ca328b57bdf6c8ebcd8504a3","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea6a89104a3d3974ca328b57bdf6c8ebcd8504a3"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-14 09:07:54.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15c62db3697ed6e460f214acdaf40757fcbb9777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15c62db3697ed6e460f214acdaf40757fcbb9777"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ea6a89104a3d3974ca328b57bdf6c8ebcd8504a3","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"87a4f619d971e7c911cff3bbeebb2f110486406e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2017-12-15 00:15:30.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/8"}}},"commit":{"parents":[{"commit":"6018ce38dcc68a9ef88d8546dc4f9b5121990dd4","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6018ce38dcc68a9ef88d8546dc4f9b5121990dd4"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-15 00:15:30.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87a4f619d971e7c911cff3bbeebb2f110486406e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87a4f619d971e7c911cff3bbeebb2f110486406e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6018ce38dcc68a9ef88d8546dc4f9b5121990dd4","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":9,"change_status":"MERGED"}],"branch":"refs/heads/master"},"234ece5cab06f2383eab49a2ea091ad2febbf7ea":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2017-12-20 22:05:40.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/9"}}},"commit":{"parents":[{"commit":"b75b66f514b20c0e59c1a7fc2ee5489816fc07cb","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b75b66f514b20c0e59c1a7fc2ee5489816fc07cb"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-20 22:05:40.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/234ece5cab06f2383eab49a2ea091ad2febbf7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/234ece5cab06f2383eab49a2ea091ad2febbf7ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b75b66f514b20c0e59c1a7fc2ee5489816fc07cb","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"383ff3f9a8c03ea7f9e0f5950ee978166dcd313b":{"kind":"REWORK","_number":10,"created":"2017-12-25 02:47:47.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/10"}}},"commit":{"parents":[{"commit":"51b1936052ff532115474370233afe251a8d57c0","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51b1936052ff532115474370233afe251a8d57c0"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-25 02:46:26.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/383ff3f9a8c03ea7f9e0f5950ee978166dcd313b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/383ff3f9a8c03ea7f9e0f5950ee978166dcd313b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"51b1936052ff532115474370233afe251a8d57c0","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b9d73ee70d08dc9c9519309f3ef73cdf77ccf98a":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2017-12-28 01:30:04.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/11"}}},"commit":{"parents":[{"commit":"b05ba8339d8c090a14abfa426c9d94338c8bbe50","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b05ba8339d8c090a14abfa426c9d94338c8bbe50"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-28 01:30:04.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9d73ee70d08dc9c9519309f3ef73cdf77ccf98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b9d73ee70d08dc9c9519309f3ef73cdf77ccf98a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b05ba8339d8c090a14abfa426c9d94338c8bbe50","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5d2b32b457652079e1eeac1fc904b0b0298130bb":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2018-01-02 12:55:03.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/12"}}},"commit":{"parents":[{"commit":"e98465a20f52b49f3fbbb7e204708786ced2d6de","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e98465a20f52b49f3fbbb7e204708786ced2d6de"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-01-02 12:55:03.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d2b32b457652079e1eeac1fc904b0b0298130bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d2b32b457652079e1eeac1fc904b0b0298130bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e98465a20f52b49f3fbbb7e204708786ced2d6de","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7d62f4f67cabde4bd87f261b0c48595bb607b7c8":{"kind":"REWORK","_number":13,"created":"2018-01-11 06:09:40.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/13"}}},"commit":{"parents":[{"commit":"2f7b3405f19fd27650996b80ae1b84dda0365807","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f7b3405f19fd27650996b80ae1b84dda0365807"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-01-11 06:09:13.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d62f4f67cabde4bd87f261b0c48595bb607b7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d62f4f67cabde4bd87f261b0c48595bb607b7c8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2f7b3405f19fd27650996b80ae1b84dda0365807","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"881b9d20da17bff100a58e487c5864cfad8cab3f":{"kind":"REWORK","_number":14,"created":"2018-01-18 04:36:38.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/14"}}},"commit":{"parents":[{"commit":"f4f0f312a65743d05fcd870f625d73f477b25a0c","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4f0f312a65743d05fcd870f625d73f477b25a0c"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-01-18 04:36:08.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/881b9d20da17bff100a58e487c5864cfad8cab3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/881b9d20da17bff100a58e487c5864cfad8cab3f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f4f0f312a65743d05fcd870f625d73f477b25a0c","is_merged_in_target_branch":false,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"14b1f5d44180e168395de1ad9596f0d0478fde59":{"kind":"REWORK","_number":15,"created":"2018-01-22 03:17:27.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/15"}}},"commit":{"parents":[{"commit":"a326b03339123d79e71edfaca3215def3756b9ae","subject":"[placement] Add sending global request ID in delete (3)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a326b03339123d79e71edfaca3215def3756b9ae"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-01-22 03:15:46.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14b1f5d44180e168395de1ad9596f0d0478fde59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/14b1f5d44180e168395de1ad9596f0d0478fde59"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a326b03339123d79e71edfaca3215def3756b9ae","is_merged_in_target_branch":true,"change_id":"I1dac3d340fe7077095d68f803cf5335ffd5b3364","change_number":526557,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ac0fccffee3c7519bb435a30c04ad6487856b1e6":{"kind":"REWORK","_number":16,"created":"2018-02-01 22:19:57.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/16"}}},"commit":{"parents":[{"commit":"d189e682277b4b5d5b70c46a2ad065e29ccccb22","subject":"Merge \"De-duplicate _numa_get_flavor_XXX_map_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d189e682277b4b5d5b70c46a2ad065e29ccccb22"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-02-01 22:19:20.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac0fccffee3c7519bb435a30c04ad6487856b1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac0fccffee3c7519bb435a30c04ad6487856b1e6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d189e682277b4b5d5b70c46a2ad065e29ccccb22","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3aed6aa60b49023a291b115cc288071949a2b905":{"kind":"REWORK","_number":17,"created":"2018-02-02 01:37:27.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/17"}}},"commit":{"parents":[{"commit":"d189e682277b4b5d5b70c46a2ad065e29ccccb22","subject":"Merge \"De-duplicate _numa_get_flavor_XXX_map_list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d189e682277b4b5d5b70c46a2ad065e29ccccb22"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-02-02 01:36:49.000000000","tz":540},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3aed6aa60b49023a291b115cc288071949a2b905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3aed6aa60b49023a291b115cc288071949a2b905"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d189e682277b4b5d5b70c46a2ad065e29ccccb22","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6079b722cbfc96dfab884a8350d8ade67e3fb840":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2018-02-04 23:40:53.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/23/526823/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/526823/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/526823/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/526823/18"}}},"commit":{"parents":[{"commit":"38eb95384f2d7af3dd39c72ecfa53eafe9daa738","subject":"Merge \"Don\u0027t wait for vif plug events during _hard_reboot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38eb95384f2d7af3dd39c72ecfa53eafe9daa738"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2017-12-09 05:04:34.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-02-04 23:40:53.000000000","tz":0},"subject":"[placement] Add sending global request ID in post","message":"[placement] Add sending global request ID in post\n\nAdd the \u0027X-Openstack-Request-Id\u0027 header\nin the request of POST in SchedulerReportClient.\nWhen creating a resource provider and creating a resource class,\nthe header is added.\n\nSubsequent patches will add the header in the other cases.\n\nChange-Id: I39d8c71432b3adf7e5bdde1c6cb6f089a9c79614\nPartial-Bug: #1734625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6079b722cbfc96dfab884a8350d8ade67e3fb840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6079b722cbfc96dfab884a8350d8ade67e3fb840"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38eb95384f2d7af3dd39c72ecfa53eafe9daa738","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
