)]}'
{"id":"openstack%2Fnova~529585","triplet_id":"openstack%2Fnova~master~I61a0bee9e71e9a67f6a7c04a7bfd6e77fe818a77","project":"openstack/nova","branch":"master","topic":"trivia","hashtags":[],"change_id":"I61a0bee9e71e9a67f6a7c04a7bfd6e77fe818a77","subject":"Remove unused argument from LibvirtDriver._disconnect_volume","status":"MERGED","created":"2017-12-21 13:28:53.000000000","updated":"2017-12-23 00:16:26.000000000","submitted":"2017-12-23 00:16:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"529585-1513988186311-9cedb2f7","meta_rev_id":"1a6e6c170e99cdc09c34e43a28a55c123de60d9d","_number":529585,"virtual_id_number":529585,"owner":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2017-12-22 09:47:49.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-22 09:56:13.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-21 21:30:12.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-22 16:59:37.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2017-12-21 17:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2017-12-22 09:44:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2017-12-23 00:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-22 15:47:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-22 15:54:30.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2017-12-22 08:58:02.000000000","_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"date":"2017-12-22 17:46:45.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2017-12-21 21:35:29.000000000","_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2017-12-22 16:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2017-12-21 22:33:35.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2017-12-22 09:07:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2017-12-22 16:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-21 15:30:43.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2017-12-21 17:30:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-21 21:30:12.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-21 21:35:29.000000000","updated_by":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-21 22:33:35.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2017-12-22 09:07:28.000000000","updated_by":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"reviewer":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"state":"REVIEWER"},{"updated":"2017-12-22 09:44:13.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2017-12-22 09:47:49.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-22 09:56:13.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-22 15:47:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-22 15:54:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-22 16:58:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2017-12-22 16:59:37.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-22 17:46:45.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2017-12-23 00:16:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e4f83bc9979d641e1915ca4560968e14c52b415","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-12-21 13:28:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96ec3696cde95f43f49eb213197dc8a8b32dd4f8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-21 14:25:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/529585/1/check/nova-out-of-tree-pvm/e8adb74 : SUCCESS in 56m 05s\n- nova-in-tree-pvm http://184.172.12.213/85/529585/1/check/nova-in-tree-pvm/a176238 : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fd396c91630fe486fa80c451517c2f30be9cef14","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-21 14:26:02.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11281 : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0bd46a68ad4291ba2a99fa31864d78584a15c763","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-21 14:32:40.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/1/check/tempest-dsvm-full-xenial/23de79e/ : FAILURE in 1h 01m 34s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/1/check/grenade-dsvm-xenial/5766e3a/ : FAILURE in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cc92f1255a9d4fe238bdef55091f3393b334d41","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-12-21 14:47:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLittle nit. Fix if you respin","accounts_in_message":[],"_revision_number":1},{"id":"82ee5f52782eb238cf1ba75da8fb8503aa540010","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-12-21 14:53:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a68569c8813e6e6305fd64c04537b6ea9278cdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 15:19:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/529585/1/check/openstack-tox-pep8/37785a8/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/85/529585/1/check/openstack-tox-py27/9e8dd23/ : FAILURE in 11m 42s\n- openstack-tox-py35 http://logs.openstack.org/85/529585/1/check/openstack-tox-py35/fd56a5e/ : FAILURE in 11m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/529585/1/check/build-openstack-sphinx-docs/f785110/html/ : SUCCESS in 5m 01s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-py35/bacf71d/ : FAILURE in 1h 35m 58s\n- openstack-tox-functional http://logs.openstack.org/85/529585/1/check/openstack-tox-functional/3158283/ : SUCCESS in 24m 01s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/529585/1/check/openstack-tox-functional-py35/a3d289a/ : SUCCESS in 18m 19s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/85/529585/1/check/legacy-grenade-dsvm-neutron-multinode/87c97d0/ : FAILURE in 1h 07m 24s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/529585/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/c5feb3d/ : FAILURE in 54m 11s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-cells/d537b1c/ : FAILURE in 57m 35s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/fb67492/ : FAILURE in 1h 25m 41s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-neutron-multinode-full/21fb493/ : FAILURE in 1h 46m 58s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-neutron-nova-next-full/104450a/ : FAILURE in 1h 23m 55s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-multinode-live-migration/05350e3/ : SUCCESS in 46m 37s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/529585/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/e02e515/ : SUCCESS in 37m 53s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/85/529585/1/check/legacy-tempest-dsvm-neutron-full/d315891/ : FAILURE in 1h 30m 14s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/85/529585/1/check/legacy-grenade-dsvm-neutron/2ea8bac/ : FAILURE in 41m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d64ab31f5b460e2a0ad7d60c7cc95ec51edde1a1","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2017-12-21 15:30:43.000000000","message":"Patch Set 1: Code-Review-1\n\nLooks like you missed the driver.py changes, maybe not added to the commit?","accounts_in_message":[],"_revision_number":1},{"id":"ffaa39ea144f0e344444a999c4cb6ff33e2924ef","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 15:39:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/1/check/check-dsvm-tempest-vz7-exe-minimal/1206e7b : SUCCESS in 50m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"0a51af5090bafa4d544472560755dba74c2011b1","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-12-21 15:46:31.000000000","message":"Patch Set 1:\n\nNope, just forgot again that we run libvirt unit tests outside libvirt unit tests.","accounts_in_message":[],"_revision_number":1},{"id":"2b3f57c9b840b859c2f8886ea29628b610542b2c","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-12-21 15:58:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6cf0cfe1a8cf7b56b8371634b97d8d9e820d411","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-21 17:08:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/529585/2/check/nova-out-of-tree-pvm/a1752fc : SUCCESS in 55m 52s\n- nova-in-tree-pvm http://184.172.12.213/85/529585/2/check/nova-in-tree-pvm/864c743 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e174302f70aa9ccd084bf50b63c7a32479cd64ef","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-21 17:30:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/tempest-dsvm-full-xenial/68d572c/ : SUCCESS in 54m 21s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/grenade-dsvm-xenial/aef0e5e/ : SUCCESS in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08bf56a8072fd011637c318330fb9af46b6b9d45","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-21 17:37:45.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11314 : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f934d1c91aa8782b70d7af43914aff3d28c413d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 20:46:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/529585/2/check/openstack-tox-pep8/fdcedd9/ : SUCCESS in 9m 23s\n- openstack-tox-py27 http://logs.openstack.org/85/529585/2/check/openstack-tox-py27/a1ffff5/ : SUCCESS in 12m 39s\n- openstack-tox-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-py35/079a39e/ : SUCCESS in 11m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/529585/2/check/build-openstack-sphinx-docs/4cee392/html/ : SUCCESS in 5m 44s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-py35/6765727/ : SUCCESS in 1h 18m 16s\n- openstack-tox-functional http://logs.openstack.org/85/529585/2/check/openstack-tox-functional/d777a52/ : SUCCESS in 20m 39s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-functional-py35/8c513fc/ : SUCCESS in 26m 48s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode/5593e34/ : SUCCESS in 49m 13s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/674ec84/ : SUCCESS in 58m 53s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-cells/4787751/ : TIMED_OUT in 4h 26m 11s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/24c46a3/ : SUCCESS in 1h 13m 14s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-multinode-full/8fa0079/ : SUCCESS in 1h 25m 10s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-nova-next-full/677373b/ : SUCCESS in 1h 17m 42s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-multinode-live-migration/65dd5fa/ : SUCCESS in 38m 57s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/529585/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f69d62a/ : SUCCESS in 52m 03s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-full/91618da/ : SUCCESS in 1h 32m 04s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron/d95ae42/ : SUCCESS in 1h 03m 56s","accounts_in_message":[],"_revision_number":2},{"id":"647137eaece988add4290de55d65dfb6dd18cb09","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 21:30:12.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/529585/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e7287c1 : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fce034ee2e2804a959e2eda7e81b6a355ae170aa","author":{"_account_id":15286,"name":"XenProject CI","email":"openstack-ci@xenproject.org","username":"XenProject-CI","tags":["SERVICE_USER"]},"date":"2017-12-21 21:35:29.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027xen: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-xen http://logs.openstack.xenproject.org/85/529585/2/check/dsvm-tempest-xen/413854d : FAILURE in 2h 09m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6d6d61c508321c6b4e8a420020be258a48f409c9","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-21 21:40:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-minimal/a945845 : SUCCESS in 57m 20s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-multinode/f19c64d : SUCCESS in 57m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"25fdd76d77f800652f14823eb5607c69869d22b6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-12-21 22:28:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e3c0589938e240159b7488524237efcaad6c37f7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-12-21 22:33:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)\n\nlooks ok","accounts_in_message":[],"_revision_number":2},{"id":"95a0405076d95128b210b56af7a0d8419638c3ce","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-21 23:11:30.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11335 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f4f69912d75b290f66b59b0e2e01d78ee6d0e280","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-21 23:14:43.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/529585/2/check/nova-out-of-tree-pvm/e205787 : SUCCESS in 45m 32s\n- nova-in-tree-pvm http://184.172.12.213/85/529585/2/check/nova-in-tree-pvm/03e752c : FAILURE in 33m 38s","accounts_in_message":[],"_revision_number":2},{"id":"066b290f2b76860990b0b6da0176af7df76dca8a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-21 23:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/tempest-dsvm-full-xenial/b0d43a7/ : SUCCESS in 55m 11s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/grenade-dsvm-xenial/46984c6/ : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7cd5fc09f2691abf512186b218075e6d745e3b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-21 23:54:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/529585/2/check/openstack-tox-pep8/474d876/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/85/529585/2/check/openstack-tox-py27/a8cd66f/ : SUCCESS in 10m 56s\n- openstack-tox-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-py35/df0c385/ : SUCCESS in 10m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/529585/2/check/build-openstack-sphinx-docs/8fee814/html/ : SUCCESS in 5m 34s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-py35/270b931/ : SUCCESS in 1h 05m 45s\n- openstack-tox-functional http://logs.openstack.org/85/529585/2/check/openstack-tox-functional/1a47c4c/ : SUCCESS in 26m 23s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-functional-py35/ed72ba4/ : SUCCESS in 21m 17s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode/9ba7a62/ : SUCCESS in 52m 22s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/2efe9c4/ : SUCCESS in 51m 18s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-cells/3cf705c/ : SUCCESS in 43m 28s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/6e222e8/ : SUCCESS in 1h 09m 47s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-multinode-full/84cbc4b/ : FAILURE in 1h 18m 42s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-nova-next-full/4b420a4/ : SUCCESS in 1h 17m 09s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-multinode-live-migration/e09d84f/ : SUCCESS in 36m 43s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/529585/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/769a38f/ : SUCCESS in 36m 52s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-full/e94a32a/ : FAILURE in 1h 11m 39s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron/ebd7500/ : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":2},{"id":"32ba10dc23d418dd4076381377fb0966662e9f65","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-22 00:25:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-minimal/172a1ab : SUCCESS in 1h 05m 03s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-multinode/4d09fce : SUCCESS in 56m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"8bdbc3ba84a22ca92322523a9cd4a580b9925d65","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-22 04:41:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/85/529585/2/check-vote/ext-nova-zuul/463700c : FAILURE in 58m 36s","accounts_in_message":[],"_revision_number":2},{"id":"90b8ee580927027ce25ed8ae54999e31c7066edd","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2017-12-22 08:58:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"945929d582875e85d56a2df56c155f30977155f2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-22 08:58:50.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"412f8c933d1045ccdc18e36eef596c9110976cf6","author":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"date":"2017-12-22 09:07:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f51f50c5743be60c7ccfb04e16805ca7ccfca3b3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2017-12-22 09:44:13.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-11364 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bbd36012a4cbe4922c33a018dc97c258ca688f4c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2017-12-22 09:47:49.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/529585/2/check/nova-out-of-tree-pvm/acf6009 : FAILURE in 48m 00s\n- nova-in-tree-pvm http://184.172.12.213/85/529585/2/check/nova-in-tree-pvm/d0bfc33 : FAILURE in 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2c89c325487877a27f6f3d66f3167e4d155040fe","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2017-12-22 09:56:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-minimal/fd09b04 : SUCCESS in 50m 00s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/85/529585/2/check/check-dsvm-tempest-vz7-exe-multinode/c61ceaf : SUCCESS in 57m 23s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"88c54fba17feb5b605ffec212f2ff544caeae046","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-22 09:56:35.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/tempest-dsvm-full-xenial/2998f14/ : SUCCESS in 57m 37s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/grenade-dsvm-xenial/b7c28fe/ : SUCCESS in 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21c0366971c011372f37db9800863c96fbe8fabc","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2017-12-22 13:41:47.000000000","message":"Patch Set 2:\n\nrecheck bug 1732199","accounts_in_message":[],"_revision_number":2},{"id":"0892a33505fa591e7156f97243e8a36b9a42f3c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-22 13:48:47.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"353d5699f91743eb09908cbbe4bb23bf3e66613a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2017-12-22 14:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/tempest-dsvm-full-xenial/607e833/ : SUCCESS in 55m 01s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/529585/2/check/grenade-dsvm-xenial/2721b2c/ : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86517aee9d75de8fe0fffe838089d922495012e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-22 15:43:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/529585/2/check/openstack-tox-pep8/ed0901a/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/85/529585/2/check/openstack-tox-py27/2c7069d/ : SUCCESS in 10m 41s\n- openstack-tox-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-py35/d7d85ef/ : SUCCESS in 11m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/529585/2/check/build-openstack-sphinx-docs/76f7b36/html/ : SUCCESS in 5m 11s\n- tempest-full http://logs.openstack.org/85/529585/2/check/tempest-full/1979d84/ : SUCCESS in 1h 52m 12s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron/2b90314/ : SUCCESS in 50m 41s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-py35/6c51dd6/ : SUCCESS in 1h 16m 44s\n- openstack-tox-functional http://logs.openstack.org/85/529585/2/check/openstack-tox-functional/8e72780/ : SUCCESS in 22m 42s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/529585/2/check/openstack-tox-functional-py35/8e94b6d/ : SUCCESS in 24m 23s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode/2f6fc61/ : SUCCESS in 50m 59s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/f5701ae/ : SUCCESS in 53m 08s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-cells/da84d35/ : SUCCESS in 40m 00s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/b5ada3d/ : SUCCESS in 1h 13m 02s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-multinode-full/963d1c7/ : SUCCESS in 1h 43m 31s (non-voting)\n- legacy-tempest-dsvm-neutron-nova-next-full http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-neutron-nova-next-full/ec7b6f7/ : SUCCESS in 1h 05m 39s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/85/529585/2/check/legacy-tempest-dsvm-multinode-live-migration/52ae034/ : SUCCESS in 35m 47s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/529585/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c1ebdcb/ : SUCCESS in 57m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a7c19ca36558069d537a264c6ebebe5561be7db","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2017-12-22 15:47:13.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"006a5b971639441c7179b5039359cbf04f4a9027","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2017-12-22 15:54:30.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-85-529585-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"495f5b2c3a00ea70bfe1682ec9f2e22ec0488aaf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2017-12-22 16:58:27.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4933b9fbbb1f13aed13915bc1e09b608e144853f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2017-12-22 16:59:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/85/529585/2/check-nova/Nova-ML2-Sriov/69710ed : FAILURE in 48m 42s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/85/529585/2/check-nova/Nova-MACVTAP-ML2-Sriov/2581857 : FAILURE in 1h 13m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"e367a472e96c7e032bf3fd38f920ddc02f8eee2b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2017-12-22 17:46:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/529585/2/ : SUCCESS in 3h 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5323c12191bc33b611aac4deda928d254ee1fb32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-22 22:48:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7da5547c76d7dfa6d158045a0c48fa835260e8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-23 00:16:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/529585/2/gate/openstack-tox-pep8/589e421/ : SUCCESS in 8m 53s\n- openstack-tox-py27 http://logs.openstack.org/85/529585/2/gate/openstack-tox-py27/fa9ee17/ : SUCCESS in 18m 01s\n- openstack-tox-py35 http://logs.openstack.org/85/529585/2/gate/openstack-tox-py35/2bfa29a/ : SUCCESS in 12m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/529585/2/gate/build-openstack-sphinx-docs/47c4760/html/ : SUCCESS in 11m 50s\n- tempest-full http://logs.openstack.org/85/529585/2/gate/tempest-full/02b6bf8/ : SUCCESS in 1h 24m 19s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/85/529585/2/gate/legacy-grenade-dsvm-neutron/b6bca7d/ : SUCCESS in 45m 15s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/85/529585/2/gate/legacy-tempest-dsvm-py35/b816d23/ : SUCCESS in 1h 15m 32s\n- openstack-tox-functional http://logs.openstack.org/85/529585/2/gate/openstack-tox-functional/69ed01b/ : SUCCESS in 20m 17s\n- openstack-tox-functional-py35 http://logs.openstack.org/85/529585/2/gate/openstack-tox-functional-py35/68b2203/ : SUCCESS in 21m 26s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/85/529585/2/gate/legacy-tempest-dsvm-cells/b5fb40a/ : SUCCESS in 35m 57s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/85/529585/2/gate/legacy-tempest-dsvm-multinode-live-migration/7564109/ : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d5fc9b3996b02365f4203a66c3d37b8d3c1537b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-23 00:16:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b912556ae01a4925aabc2610756a7564e32df637","revisions":{"ac1239c301ba2a73e59a5ffb99f866df33d0f5b6":{"kind":"REWORK","_number":1,"created":"2017-12-21 13:28:53.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/85/529585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/529585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/529585/1"}}},"commit":{"parents":[{"commit":"cfa6d3ae30282991f66606af1e06613698b0efb3","subject":"Removed unused argument from LibvirtDriver._connect_volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfa6d3ae30282991f66606af1e06613698b0efb3"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2017-12-17 04:17:22.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2017-12-17 04:24:01.000000000","tz":0},"subject":"Remove unused argument from LibvirtDriver._disconnect_volume","message":"Remove unused argument from LibvirtDriver._disconnect_volume\n\n_disconnect_volume was being passed disk_dev, which was is name of the\ndisk as presented to the gust. However, _disconnect_volume is only\nconcerned with unmounting the volume from the host, so this isn\u0027t\nrelevant.\n\nA couple of volume drivers were using it for logging. We remove these\nbecause it wasn\u0027t done consistently, is better done by the caller, and\nisn\u0027t required as the information is available from other log\nmessages.\n\nIn the very minor associated debug log cleanup we also remove logging\nof connection_info from one volume driver, which is potential security\nbug.\n\nWe also do some associated cleanup in a few volume driver tests which\nwere assuming that disconnect_volume was being passed disk_info rather\nthan disk_dev, and that connection_info[\u0027data\u0027] is related to\ndisk_info, which it isn\u0027t.\n\nChange-Id: I61a0bee9e71e9a67f6a7c04a7bfd6e77fe818a77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac1239c301ba2a73e59a5ffb99f866df33d0f5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac1239c301ba2a73e59a5ffb99f866df33d0f5b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cfa6d3ae30282991f66606af1e06613698b0efb3","is_merged_in_target_branch":true,"change_id":"I843b5c46f9f93a30e7121259feff17a8170a2e48","change_number":529556,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b912556ae01a4925aabc2610756a7564e32df637":{"kind":"REWORK","_number":2,"created":"2017-12-21 15:58:45.000000000","uploader":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"ref":"refs/changes/85/529585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/529585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/529585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/529585/2"}}},"commit":{"parents":[{"commit":"cfa6d3ae30282991f66606af1e06613698b0efb3","subject":"Removed unused argument from LibvirtDriver._connect_volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfa6d3ae30282991f66606af1e06613698b0efb3"}]}],"author":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2017-12-17 04:17:22.000000000","tz":0},"committer":{"name":"Matthew Booth","email":"mbooth@redhat.com","date":"2017-12-17 07:12:04.000000000","tz":0},"subject":"Remove unused argument from LibvirtDriver._disconnect_volume","message":"Remove unused argument from LibvirtDriver._disconnect_volume\n\n_disconnect_volume was being passed disk_dev, which was is name of the\ndisk as presented to the gust. However, _disconnect_volume is only\nconcerned with unmounting the volume from the host, so this isn\u0027t\nrelevant.\n\nA couple of volume drivers were using it for logging. We remove these\nbecause it wasn\u0027t done consistently, is better done by the caller, and\nisn\u0027t required as the information is available from other log\nmessages.\n\nIn the very minor associated debug log cleanup we also remove logging\nof connection_info from one volume driver, which is potential security\nbug.\n\nWe also do some associated cleanup in a few volume driver tests which\nwere assuming that disconnect_volume was being passed disk_info rather\nthan disk_dev, and that connection_info[\u0027data\u0027] is related to\ndisk_info, which it isn\u0027t.\n\nChange-Id: I61a0bee9e71e9a67f6a7c04a7bfd6e77fe818a77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b912556ae01a4925aabc2610756a7564e32df637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b912556ae01a4925aabc2610756a7564e32df637"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cfa6d3ae30282991f66606af1e06613698b0efb3","is_merged_in_target_branch":true,"change_id":"I843b5c46f9f93a30e7121259feff17a8170a2e48","change_number":529556,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
