)]}'
{"id":"openstack%2Fnova~539300","triplet_id":"openstack%2Fnova~master~Ia83af4dfcc0c040679b0d0cd5282830fca27bd63","project":"openstack/nova","branch":"master","topic":"bp/request-traits-in-nova","hashtags":[],"change_id":"Ia83af4dfcc0c040679b0d0cd5282830fca27bd63","subject":"Mention required traits in the flavors user docs","status":"MERGED","created":"2018-01-30 19:11:37.000000000","updated":"2018-02-01 21:25:54.000000000","submitted":"2018-02-01 21:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"539300-1517520354765-efdaf465","meta_rev_id":"595907509fcf28181ae6e91b05bf71970c570928","_number":539300,"virtual_id_number":539300,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-01-31 17:24:32.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-01-31 23:53:57.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2018-01-31 17:52:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2018-02-01 21:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-01-31 16:07:13.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-01-31 16:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2018-02-01 09:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-31 17:13:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-31 07:39:56.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2018-01-31 17:13:29.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-01-31 17:24:32.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-31 17:52:59.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-01-31 23:53:57.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-02-01 09:12:09.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2018-02-01 21:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e1b7e96900ccc3accaef07e7b8d9b2dace6e09b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-30 19:11:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"791403fdfa042544633b1b79635ddd8e8ff05442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 20:01:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/539300/1/check/openstack-tox-pep8/840d021/ : SUCCESS in 10m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/539300/1/check/build-openstack-sphinx-docs/d65b501/html/ : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b99a14d4ed3d46f675045e902b3d6787ee18b6f5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-30 20:32:35.000000000","message":"Patch Set 1:\n\nhttp://logs.openstack.org/00/539300/1/check/build-openstack-sphinx-docs/d65b501/html/user/flavors.html#extra-specs","accounts_in_message":[],"_revision_number":1},{"id":"9f3e04f44265a188589bacea7f7b32af022a5495","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-30 20:59:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/00/539300/1/check/nova-out-of-tree-pvm/5fabd48 : SUCCESS in 48m 35s\n- nova-in-tree-pvm http://184.172.12.213/00/539300/1/check/nova-in-tree-pvm/a14d1e2 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0c4302c8fde836670a7e9e9012592ffe60e4e758","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-30 20:59:26.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13336 : SUCCESS in 1h 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"14190fb5f6a534338b5aea042428ed61abcac216","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-31 02:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/539300/1/ : SUCCESS in 2h 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a614c396c5dd8e0a2fb28626d65dcf4f83fdc4ed","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-01-31 07:39:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bce71b3b0bd4bfbc9a8712c51f07e774d85327d7","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-01-31 08:17:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"023f42f820bd15eec0e93351edefe84d7b0e141b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 08:43:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/539300/2/check/openstack-tox-pep8/682a02d/ : SUCCESS in 8m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/539300/2/check/build-openstack-sphinx-docs/c33f405/html/ : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b5d1b4e15af76271d0ac8ae97e50c757008674c7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-31 08:50:19.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2\n\nLGTM!\nCheers,\ngibi","accounts_in_message":[],"_revision_number":2},{"id":"0da6d11572856348f538bae28aa89478eef24370","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-31 09:09:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/00/539300/2/check/nova-out-of-tree-pvm/b78c00d : SUCCESS in 48m 58s\n- nova-in-tree-pvm http://184.172.12.213/00/539300/2/check/nova-in-tree-pvm/ce7b163 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":2},{"id":"eaf92f3ecf2b140c977822d749fd50c30b4f4275","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-31 10:44:52.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13382 : SUCCESS in 2h 24m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a96b692d8c57e733158a8bcfa69b881fc62cfec8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-31 11:08:00.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/539300/2/ : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f8131e632f584437a41d2217c990a82ab89b9167","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 15:01:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"862fe3de0ebd4f25b226c3c65e018c54bfc2ccb8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 16:07:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bc31ba62955fd66af2217e500d228037a71350e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 16:41:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/539300/3/check/openstack-tox-pep8/701d366/ : SUCCESS in 7m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/539300/3/check/build-openstack-sphinx-docs/aea617e/html/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8fc3a9df37a5423b22599e62a1e00f4b21ee3fc1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-01-31 17:13:29.000000000","message":"Patch Set 3: Workflow+1\n\ntrivial rebase.","accounts_in_message":[],"_revision_number":3},{"id":"acb5e766e7fe6c100998f4bb7a3dec657c1d9b5e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-31 17:24:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/00/539300/3/check/nova-out-of-tree-pvm/9dc39c6 : SUCCESS in 1h 07m 10s\n- nova-in-tree-pvm http://184.172.12.213/00/539300/3/check/nova-in-tree-pvm/7c2d972 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":3},{"id":"82445284cc65ce889f9f732605dbe4949a48998c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-31 17:52:59.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13426 : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":3},{"id":"109b092f874d267473ff381bed918fbe19a8a6c9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-01-31 23:53:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/539300/3/ : SUCCESS in 3h 27m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8fef48a8f19f763a544fd00901475705b2ec208e","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2018-02-01 09:12:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e399ff0dd43397bc211888da02146a6bbab91368","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 09:37:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"74279ac5c96d5d00b5209ed0c83752ca38a7761b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 21:25:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/00/539300/3/gate/openstack-tox-pep8/abd1363/ : SUCCESS in 9m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/539300/3/gate/build-openstack-sphinx-docs/57cfae4/html/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2f2b6eff19f47c6e686b20d912998ca5ee5693ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 21:25:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e9a631ec8cc35863af069096c9b81ec5a09b4bd","revisions":{"c310bea1eb833c1b6b03f8788bbf168a1f961817":{"kind":"REWORK","_number":1,"created":"2018-01-30 19:11:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/00/539300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/539300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/539300/1"}}},"commit":{"parents":[{"commit":"0788f422297282b711432a576ea7aca263c4ee2b","subject":"Fix nits in support traits changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0788f422297282b711432a576ea7aca263c4ee2b"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-30 19:10:50.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-30 19:10:50.000000000","tz":-300},"subject":"Mention required traits in the flavors user docs","message":"Mention required traits in the flavors user docs\n\nThis takes most of the release note and adds it to the user\nflavor docs which is more discoverable for an end user.\n\nChange-Id: Ia83af4dfcc0c040679b0d0cd5282830fca27bd63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c310bea1eb833c1b6b03f8788bbf168a1f961817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c310bea1eb833c1b6b03f8788bbf168a1f961817"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0788f422297282b711432a576ea7aca263c4ee2b","is_merged_in_target_branch":false,"change_id":"I366b97ef3c141834f48949700edb968a7c7c4167","change_number":537351,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"277f7ce5e853cef76c83270ab4631d65176c8bc2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-01-31 08:17:04.000000000","uploader":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"ref":"refs/changes/00/539300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/539300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/539300/2"}}},"commit":{"parents":[{"commit":"2d413f8033ad78408587824756380caf8030fcc7","subject":"Fix nits in support traits changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d413f8033ad78408587824756380caf8030fcc7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-30 19:10:50.000000000","tz":-300},"committer":{"name":"He Jie Xu","email":"hejie.xu@intel.com","date":"2018-01-31 08:43:40.000000000","tz":480},"subject":"Mention required traits in the flavors user docs","message":"Mention required traits in the flavors user docs\n\nThis takes most of the release note and adds it to the user\nflavor docs which is more discoverable for an end user.\n\nChange-Id: Ia83af4dfcc0c040679b0d0cd5282830fca27bd63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/277f7ce5e853cef76c83270ab4631d65176c8bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/277f7ce5e853cef76c83270ab4631d65176c8bc2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2d413f8033ad78408587824756380caf8030fcc7","is_merged_in_target_branch":false,"change_id":"I366b97ef3c141834f48949700edb968a7c7c4167","change_number":537351,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3e9a631ec8cc35863af069096c9b81ec5a09b4bd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-01-31 16:07:13.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/00/539300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/539300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/539300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/539300/3"}}},"commit":{"parents":[{"commit":"284ba35c33053f9ef2786eacf81d6c8cdcf8e24a","subject":"Fix nits in support traits changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/284ba35c33053f9ef2786eacf81d6c8cdcf8e24a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-30 19:10:50.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-01-31 16:07:12.000000000","tz":-300},"subject":"Mention required traits in the flavors user docs","message":"Mention required traits in the flavors user docs\n\nThis takes most of the release note and adds it to the user\nflavor docs which is more discoverable for an end user.\n\nChange-Id: Ia83af4dfcc0c040679b0d0cd5282830fca27bd63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e9a631ec8cc35863af069096c9b81ec5a09b4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e9a631ec8cc35863af069096c9b81ec5a09b4bd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"284ba35c33053f9ef2786eacf81d6c8cdcf8e24a","is_merged_in_target_branch":true,"change_id":"I366b97ef3c141834f48949700edb968a7c7c4167","change_number":537351,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
