)]}'
{"id":"openstack%2Fnova~539654","triplet_id":"openstack%2Fnova~master~Ibcb489efb364c03167b7398f615c9a15f6867370","project":"openstack/nova","branch":"master","topic":"bug/1746531","hashtags":[],"change_id":"Ibcb489efb364c03167b7398f615c9a15f6867370","subject":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.","status":"MERGED","created":"2018-01-31 20:02:00.000000000","updated":"2018-02-01 21:26:11.000000000","submitted":"2018-02-01 21:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"539654-1517520371030-021d5a3d","meta_rev_id":"255fa0773df2f04c3ba11aeb5de318dbdc9077e3","_number":539654,"virtual_id_number":539654,"owner":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-01-31 22:42:04.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-01 05:25:08.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2018-01-31 23:02:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-02-01 21:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2018-02-01 09:35:02.000000000","_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2018-02-01 09:33:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":2,"date":"2018-01-31 21:49:14.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-02-01 09:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-02-01 09:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},{"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-31 21:13:12.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2018-01-31 21:49:14.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-01-31 22:42:04.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-31 23:02:16.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-02-01 05:25:08.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-02-01 09:30:33.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-02-01 09:33:47.000000000","updated_by":{"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},"reviewer":{"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},"state":"REVIEWER"},{"updated":"2018-02-01 21:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a00770a3dde65c8c93f5a5000ca3f18ba06a2688","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-01-31 20:02:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7ea7f73fffa49b66fceec58ce034dadcf3e31fb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 20:08:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ae068a97e3f58785037bd6706a979267df534969","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 20:30:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/54/539654/1/check/openstack-tox-pep8/71efebd/ : SUCCESS in 8m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/539654/1/check/build-openstack-sphinx-docs/668aa6c/ : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"22f0ceec403ba00227c8d222197a1b4873bdee34","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-01-31 20:57:41.000000000","message":"Patch Set 1:\n\n\u003e (4 comments)\n\n@Matt: Thank for review, what about zuul failure ? is that something I have to fix ?\nor just upload new patch ?","accounts_in_message":[],"_revision_number":1},{"id":"ff4474f9119f944f8dcb755c610d74e79bcff1f0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-31 20:58:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/54/539654/1/check/nova-out-of-tree-pvm/ee46a15 : SUCCESS in 55m 31s\n- nova-in-tree-pvm http://184.172.12.213/54/539654/1/check/nova-in-tree-pvm/d04d1c2 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0979abe424dd808f4cf24527750eb378ae6b3e29","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2018-01-31 21:13:12.000000000","message":"Patch Set 1: Code-Review-1\n\nThe same change of adding \"--until-complete\" is been done here I think ? : https://review.openstack.org/#/c/507486/29/doc/source/cli/nova-manage.rst","accounts_in_message":[],"_revision_number":1},{"id":"2076a6cf8656571363c67c8bb1dac5ea5dd2e619","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-01-31 21:38:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd7f0397da7d0925053be782cdc26515e50069f0","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-01-31 21:41:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6305696ccccd910e2632fd02992e78d6e869f466","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 21:43:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"99768d77256dbdca3aa539f759bead63bbaccf87","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-01-31 21:46:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf7463b7c5f1b9ad50c16dda025ebcd6c4dc8cc3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-01-31 21:49:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bdd1f0179d68f542a5db09d768acaab081deae40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 22:20:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/539654/4/check/openstack-tox-pep8/f1333de/ : SUCCESS in 13m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/539654/4/check/build-openstack-sphinx-docs/6e39019/html/ : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"66a58ffc595bb810f18da96cd634226230fdfd65","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-01-31 22:42:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/54/539654/4/check/nova-out-of-tree-pvm/d3f1258 : SUCCESS in 55m 26s\n- nova-in-tree-pvm http://184.172.12.213/54/539654/4/check/nova-in-tree-pvm/f792f57 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":4},{"id":"b1161210ec793d73acaf26abd0d90085ad03c2df","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-01-31 23:02:16.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-13457 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":4},{"id":"3b32c0ff212cd3d4ec3dafd4e82e55c9bb0a878a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-02-01 05:25:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/539654/4/ : SUCCESS in 3h 33m 33s","accounts_in_message":[],"_revision_number":4},{"id":"26e8a26ca6fc757bb0e3cc39d5e7d73bffc6184d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-02-01 09:30:33.000000000","message":"Patch Set 4: Workflow+1 Code-Review+2\n\nLGTM!\nCheers,\ngibi","accounts_in_message":[],"_revision_number":4},{"id":"1059af6f87a3cb122e7949f61e263431f29c0b18","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-02-01 09:33:25.000000000","message":"Patch Set 4:\n\n\u003e LGTM!\n \u003e Cheers,\n \u003e gibi\n\nThanks !","accounts_in_message":[],"_revision_number":4},{"id":"ab3c2d4d2f3ffad2d4ab07c1004ea3fedd56fd84","author":{"_account_id":27590,"name":"wangjianghui","email":"wjh975829659@163.com","username":"wangjianghui"},"date":"2018-02-01 09:33:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"51118c869964f1335e899a8967ca420a888a3ae7","author":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"date":"2018-02-01 09:35:02.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review+1\n\nThanks!!","accounts_in_message":[],"_revision_number":4},{"id":"a97eb322bf31a9ff732155e53f603bd1ff9f4476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 10:08:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d890b0bdc8b8cfdd60e6ee952d9247bff79436af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 21:26:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/539654/4/gate/openstack-tox-pep8/a329695/ : SUCCESS in 9m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/539654/4/gate/build-openstack-sphinx-docs/6854f95/html/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"217001e7f7a60b83f9acc948f15bb95423e5eb10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 21:26:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eba33a0e2d8aace6fbc889026fb80dc2c208bc8d","revisions":{"648906c84b0777d3b6c4513483adcf232580cfde":{"kind":"REWORK","_number":1,"created":"2018-01-31 20:02:00.000000000","uploader":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"ref":"refs/changes/54/539654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/539654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/539654/1"}}},"commit":{"parents":[{"commit":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","subject":"Merge \"Remove the inherits parameter for the Resource object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf"}]}],"author":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 19:58:03.000000000","tz":-300},"committer":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 19:59:10.000000000","tz":-300},"subject":"add \"--until-complete\" option for nova-manage db archive_deleted_rows man page","message":"add \"--until-complete\" option for nova-manage\ndb archive_deleted_rows man page\n\nChange-Id: Ibcb489efb364c03167b7398f615c9a15f6867370\nCloses-bug: #1746531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/648906c84b0777d3b6c4513483adcf232580cfde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/648906c84b0777d3b6c4513483adcf232580cfde"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"fa9d59d48d341653e8a8bcf6b02686bb11c04f41":{"kind":"REWORK","_number":2,"created":"2018-01-31 21:38:15.000000000","uploader":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"ref":"refs/changes/54/539654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/539654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/539654/2"}}},"commit":{"parents":[{"commit":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","subject":"Merge \"Remove the inherits parameter for the Resource object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf"}]}],"author":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 19:58:03.000000000","tz":-300},"committer":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 21:37:28.000000000","tz":-300},"subject":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.","message":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.\n\nChange-Id: Ibcb489efb364c03167b7398f615c9a15f6867370\nCloses-bug: #1746531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa9d59d48d341653e8a8bcf6b02686bb11c04f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa9d59d48d341653e8a8bcf6b02686bb11c04f41"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f541f8157874bcd424fc73f27789ab9cfea4da2e":{"kind":"REWORK","_number":3,"created":"2018-01-31 21:41:20.000000000","uploader":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"ref":"refs/changes/54/539654/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/539654/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/539654/3"}}},"commit":{"parents":[{"commit":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","subject":"Merge \"Remove the inherits parameter for the Resource object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf"}]}],"author":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 19:58:03.000000000","tz":-300},"committer":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 21:40:29.000000000","tz":-300},"subject":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.","message":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.\n\nChange-Id: Ibcb489efb364c03167b7398f615c9a15f6867370\nCloses-bug: #1746531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f541f8157874bcd424fc73f27789ab9cfea4da2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f541f8157874bcd424fc73f27789ab9cfea4da2e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"eba33a0e2d8aace6fbc889026fb80dc2c208bc8d":{"kind":"REWORK","_number":4,"created":"2018-01-31 21:46:06.000000000","uploader":{"_account_id":27277,"name":"Ameed Ashour","email":"ameed.ashour.ext@nokia.com","username":"ameeda"},"ref":"refs/changes/54/539654/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/539654/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/539654/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/539654/4"}}},"commit":{"parents":[{"commit":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","subject":"Merge \"Remove the inherits parameter for the Resource object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf"}]}],"author":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 19:58:03.000000000","tz":-300},"committer":{"name":"Ameed Ashour","email":"ameed.ashour@exalt.ps","date":"2018-01-31 21:45:24.000000000","tz":-300},"subject":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.","message":"add \"--until-complete\" option for nova-manage db archive_deleted_rows.\n\nChange-Id: Ibcb489efb364c03167b7398f615c9a15f6867370\nCloses-bug: #1746531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eba33a0e2d8aace6fbc889026fb80dc2c208bc8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eba33a0e2d8aace6fbc889026fb80dc2c208bc8d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1442d3c8cf9ab8a3cc6fe7e169c71e39abe1faf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
