)]}'
{"id":"openstack%2Fnova~549737","triplet_id":"openstack%2Fnova~master~I696a8ea8adbe9481e11407ecafd5e47b2bd29804","project":"openstack/nova","branch":"master","topic":"bug/1753443","hashtags":[],"change_id":"I696a8ea8adbe9481e11407ecafd5e47b2bd29804","subject":"Fix version cap when no nova-compute started","status":"MERGED","created":"2018-03-05 10:59:08.000000000","updated":"2018-03-07 18:25:29.000000000","submitted":"2018-03-07 17:26:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"549737-1520443608538-1e97271c","meta_rev_id":"780b276ec3d09cca98f29b15e9608b301d44e191","_number":549737,"virtual_id_number":549737,"owner":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-03-06 22:59:44.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-07 08:02:59.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"date":"2018-03-07 14:11:47.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-06 22:21:08.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-06 21:38:38.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2018-03-06 21:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2018-03-07 14:02:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-03-07 17:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-07 10:35:25.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-07 15:54:42.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-03-07 09:00:46.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2018-03-07 13:09:53.000000000","_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},{"value":0,"date":"2018-03-07 16:53:08.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2018-03-06 20:51:18.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":2,"date":"2018-03-06 20:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-03-06 21:24:20.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-03-06 21:24:20.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-06 20:36:35.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2018-03-06 20:51:18.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2018-03-06 21:15:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-06 21:24:20.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-03-06 21:38:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-06 22:21:08.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-06 22:59:44.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 08:02:59.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2018-03-07 09:00:46.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-03-07 10:35:25.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 14:02:29.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-03-07 14:11:47.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 15:54:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 16:53:08.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-03-07 17:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c2f0d2de8aec63059ffe1b601a2f9f8bc7da87f","author":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"date":"2018-03-05 10:59:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d0c2a771c77b91ff8a3f34c5448917746764179","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-05 11:47:47.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14748 : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"490f0c8bdeda56ca8b86d0d57ba170865c0de0a8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-03-05 12:03:59.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/37/549737/1/check/nova-out-of-tree-pvm/f9df54e : FAILURE in 1h 04m 12s\n- nova-in-tree-pvm http://184.172.12.213/37/549737/1/check/nova-in-tree-pvm/9230f18 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":1},{"id":"02ec78335ff4a18a138e6b013d5f2dffd14a9196","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-03-05 12:05:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/549737/1/check/check-dsvm-tempest-vz7-exe-minimal/5d9a8b2 : SUCCESS in 57m 38s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"4d0a21677dcf4ada36fad70231334390293800ee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-05 12:08:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/1/check/tempest-dsvm-full-xenial/413e140/ : SUCCESS in 1h 08m 53s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/1/check/grenade-dsvm-xenial/68464fd/ : SUCCESS in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1055f9c904ab677e1906b13d6c574b6b7c7a9417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-05 12:23:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/1/check/openstack-tox-pep8/b5a3aaf/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/1/check/openstack-tox-py27/929a5fa/ : SUCCESS in 12m 28s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/1/check/openstack-tox-py35/3861828/ : SUCCESS in 10m 46s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/1/check/build-openstack-sphinx-docs/d787ca4/html/ : SUCCESS in 5m 21s\n- tempest-full http://logs.openstack.org/37/549737/1/check/tempest-full/7dd6365/ : SUCCESS in 1h 22m 57s\n- neutron-grenade http://logs.openstack.org/37/549737/1/check/neutron-grenade/7bcf9c6/ : SUCCESS in 40m 57s\n- tempest-full-py3 http://logs.openstack.org/37/549737/1/check/tempest-full-py3/1f09052/ : SUCCESS in 1h 07m 17s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/1/check/neutron-grenade-multinode/3017aca/ : SUCCESS in 56m 14s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/be65b4c/ : SUCCESS in 53m 09s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-cells/a55faac/ : SUCCESS in 39m 26s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/f528743/ : SUCCESS in 1h 01m 21s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/1/check/neutron-tempest-multinode-full/09c4c14/ : FAILURE in 1h 11m 17s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-multinode-live-migration/8a25adc/ : SUCCESS in 57m 49s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/bd79cae/ : SUCCESS in 41m 48s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/1/check/nova-multiattach/fd7bf6e/ : SUCCESS in 49m 50s\n- nova-next http://logs.openstack.org/37/549737/1/check/nova-next/ccf509f/ : SUCCESS in 1h 15m 01s (non-voting)\n- nova-tox-functional http://logs.openstack.org/37/549737/1/check/nova-tox-functional/83656e0/ : SUCCESS in 21m 59s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/1/check/nova-tox-functional-py35/333c898/ : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5890d533655e67554b0373a6b8e58958f503cae7","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-03-05 12:43:45.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/549737/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/aac11b1 : FAILURE in 1h 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8d6d1f85d5682bc9ff2813a98e42345a9ff4461d","author":{"_account_id":11307,"name":"Andreas Scheuring","email":"andreas.scheuring@de.ibm.com","username":"uniqueName"},"date":"2018-03-05 13:01:56.000000000","message":"Patch Set 1:\n\nrecheck-zkvm","accounts_in_message":[],"_revision_number":1},{"id":"17a54086679973e260d38d03692465459a0f78a3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-05 14:01:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/1/check/tempest-dsvm-full-xenial/a370b70/ : SUCCESS in 58m 30s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/1/check/grenade-dsvm-xenial/3ccd587/ : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1b395b32018bb832f10d1fae0af951c37109337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-05 14:24:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/1/check/openstack-tox-pep8/d6ca818/ : SUCCESS in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/1/check/openstack-tox-py27/b83cfc4/ : SUCCESS in 12m 21s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/1/check/openstack-tox-py35/a605904/ : SUCCESS in 10m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/1/check/build-openstack-sphinx-docs/096d71e/html/ : SUCCESS in 5m 39s\n- tempest-full http://logs.openstack.org/37/549737/1/check/tempest-full/b4b76da/ : SUCCESS in 1h 20m 09s\n- neutron-grenade http://logs.openstack.org/37/549737/1/check/neutron-grenade/67d5221/ : SUCCESS in 51m 26s\n- tempest-full-py3 http://logs.openstack.org/37/549737/1/check/tempest-full-py3/8d86931/ : SUCCESS in 1h 06m 43s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/1/check/neutron-grenade-multinode/59d8b5e/ : SUCCESS in 52m 48s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/09ea54c/ : SUCCESS in 1h 00m 20s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-cells/0105afc/ : SUCCESS in 46m 17s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/7f4997e/ : SUCCESS in 1h 05m 27s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/1/check/neutron-tempest-multinode-full/793b819/ : FAILURE in 1h 12m 28s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/1/check/legacy-tempest-dsvm-multinode-live-migration/bac0364/ : SUCCESS in 41m 13s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c51bc96/ : SUCCESS in 38m 42s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/1/check/nova-multiattach/cb87f44/ : SUCCESS in 47m 00s\n- nova-tox-functional http://logs.openstack.org/37/549737/1/check/nova-tox-functional/fb7463e/ : SUCCESS in 23m 57s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/1/check/nova-tox-functional-py35/98ac6ac/ : SUCCESS in 23m 10s\n- nova-next http://logs.openstack.org/37/549737/1/check/nova-next/9d58f01/ : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8e334e5ad701e2cceb7828d0f07665064ab60fe","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-03-05 14:29:25.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/549737/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/4338197 : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6da397faa2fca99505c9fda08f6e798f8c85dd3d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-03-05 15:19:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/549737/1/ : FAILURE in 4h 19m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cddee233d852efa95afc7ab58a226d6f271f7c73","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-05 15:34:34.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/37/549737/1/check-vote/ext-nova-zuul/be786a2 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":1},{"id":"44fbd0dfffdd2ddef7d640744adb8280d3888dba","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 08:29:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-05/549737/1/check/tempest-dsvm-intel-nfv-xenial/f0c71da : SUCCESS in 27m 45s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-05/549737/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/7ed27e5 : SUCCESS in 1h 00m 47s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-05/549737/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/3a963a4 : SUCCESS in 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26a5ee89bddeb355babab82cb14a59fc76ade98e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-06 15:09:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"83b9b08d1c2aaf5bb33c5bdce56bde00b14aa457","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-06 15:12:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6601ef24c0a9791dc6d0af7f8df7c2ddcf68d9a0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-06 15:15:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c361f79d26a5300c10f3fec672dc5164bb9175ea","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-06 15:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac67c9b108cf57908da9ea27a1e9d8fbe4e494c","author":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"date":"2018-03-06 17:37:58.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c05a28df13f22a602d713612905b5f975fff3cab","author":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"date":"2018-03-06 17:38:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f14c04ad8928ede5cd75125302c727456ea4249","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-06 18:41:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/2/check/tempest-dsvm-full-xenial/cafb84a/ : SUCCESS in 1h 02m 31s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/2/check/grenade-dsvm-xenial/ce24d4e/ : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6432bb62a5f74be56b81cabb9c2e677d7cb24576","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-06 18:42:46.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14797 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"27c40b65b5018bd25119669db7bde051c9b7199a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 18:53:10.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/549737/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/8df61d2 : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":2},{"id":"758e1f579b93a71f5c581ca887183647550fd66b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 19:10:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-06/549737/2/check/tempest-dsvm-intel-nfv-xenial/623efac : SUCCESS in 29m 16s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-06/549737/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/d5f3980 : SUCCESS in 1h 04m 15s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-06/549737/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/4cc8c79 : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d54f00f2177194244f54971d898f44273486ecc","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-03-06 19:10:36.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/37/549737/2/check/nova-out-of-tree-pvm/e3519f4 : SUCCESS in 55m 41s\n- nova-in-tree-pvm http://184.172.12.213/37/549737/2/check/nova-in-tree-pvm/fd240ac : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":2},{"id":"86d3607290e2639b356a9f7d2666e9c2ee20d677","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 19:20:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/549737/2/check/check-dsvm-tempest-vz7-exe-minimal/771575d : SUCCESS in 55m 24s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"1a22dbb3a40b93519b7313ea04510f7ce57c3097","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-03-06 19:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/37/549737/2/check-nova/Nova-ML2-Sriov/473e8be : SUCCESS in 57m 21s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/37/549737/2/check-nova/Nova-MACVTAP-ML2-Sriov/209b63e : SUCCESS in 44m 17s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"fb41957bf5c8825329ec5dcc57dad087f9c92e96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-06 19:40:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/2/check/openstack-tox-pep8/c710e44/ : SUCCESS in 8m 08s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/2/check/openstack-tox-py27/b0401af/ : SUCCESS in 12m 15s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/2/check/openstack-tox-py35/394a8b4/ : SUCCESS in 12m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/2/check/build-openstack-sphinx-docs/9f0a2fc/html/ : SUCCESS in 5m 27s\n- tempest-full http://logs.openstack.org/37/549737/2/check/tempest-full/117fbe5/ : SUCCESS in 1h 40m 02s\n- neutron-grenade http://logs.openstack.org/37/549737/2/check/neutron-grenade/4eae87e/ : SUCCESS in 44m 02s\n- tempest-full-py3 http://logs.openstack.org/37/549737/2/check/tempest-full-py3/4a89a76/ : SUCCESS in 1h 06m 57s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/2/check/neutron-grenade-multinode/dc4eaf5/ : SUCCESS in 1h 04m 03s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/36d2df0/ : SUCCESS in 54m 57s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/2/check/legacy-tempest-dsvm-cells/f81f3dd/ : SUCCESS in 45m 25s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/fe41715/ : SUCCESS in 1h 09m 42s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/2/check/neutron-tempest-multinode-full/71b7b20/ : SUCCESS in 1h 35m 29s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/2/check/legacy-tempest-dsvm-multinode-live-migration/763af79/ : SUCCESS in 39m 36s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a9e56dd/ : SUCCESS in 45m 29s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/2/check/nova-multiattach/55de75f/ : SUCCESS in 1h 00m 46s\n- nova-tox-functional http://logs.openstack.org/37/549737/2/check/nova-tox-functional/a82883c/ : SUCCESS in 21m 17s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/2/check/nova-tox-functional-py35/ce79588/ : SUCCESS in 24m 06s\n- nova-next http://logs.openstack.org/37/549737/2/check/nova-next/8442157/ : SUCCESS in 1h 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ae2326183782f4e4987794136e6db37a9eb7cf2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-06 19:44:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nOne more thing and then I\u0027m good.","accounts_in_message":[],"_revision_number":2},{"id":"0c7be7b79ad56e39d002a250c927d70e30713431","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-03-06 19:59:43.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-549737-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"9f2eff292b54b59b288a49a965f653acf4626a6d","author":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"date":"2018-03-06 20:15:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e5300fbaa1d2d971ba33b7669e26889c58ab3fb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-06 20:36:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b24d3cd2f6dd18c61f325f84cecee4e774ddb0e3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-06 20:40:14.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks OK. I\u0027m testing here first:\n\nhttps://review.openstack.org/#/c/550108/\n\nIf that\u0027s happy I\u0027ll then +W.","accounts_in_message":[],"_revision_number":3},{"id":"878c433e0083ee0d3bfdc30716f37e41d8476c1d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-03-06 20:51:18.000000000","message":"Patch Set 3: Code-Review+2\n\nYeah, this looks spot-on to me. Thanks!\n\n-jay","accounts_in_message":[],"_revision_number":3},{"id":"6269872d8805e4c79cd1b49b9245d20690dff277","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-06 21:07:54.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14807 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":3},{"id":"c910e52ef6c6682600922c4481ce389c9db27b08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-06 21:15:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/tempest-dsvm-full-xenial/8de0d36/ : SUCCESS in 58m 32s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/grenade-dsvm-xenial/92d19c3/ : SUCCESS in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c9501bd1f4604048eec2863c61dbdd3d592256e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-06 21:24:20.000000000","message":"Patch Set 3: Workflow+1 Code-Review+2\n\ntempest-full is passing on https://review.openstack.org/#/c/550108/ now so this looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"9005b863b949318300c861d9f525e5e8812ec938","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-03-06 21:38:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://52.169.200.208/37/549737/3/check-nova/Nova-ML2-Sriov/1f246b9 : SUCCESS in 1h 01m 25s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://52.169.200.208/37/549737/3/check-nova/Nova-MACVTAP-ML2-Sriov/6b29855 : SUCCESS in 48m 58s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"612635b2805b4273ef964490a0d22deec99ce036","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 21:42:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-06/549737/3/check/tempest-dsvm-intel-nfv-xenial/9740472 : SUCCESS in 30m 48s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-06/549737/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/16f1775 : SUCCESS in 1h 02m 40s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-06/549737/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0ad2f3a : SUCCESS in 54m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"559361445648b2efdb6c8e1f2464b981b90cde97","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 21:45:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/549737/3/check/check-dsvm-tempest-vz7-exe-minimal/1bd1466 : SUCCESS in 1h 28m 14s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"666bcf65223785c4af80a9824ccac384a16cafb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-06 21:56:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/check/openstack-tox-pep8/3cf8574/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/check/openstack-tox-py27/80740a2/ : SUCCESS in 12m 09s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/check/openstack-tox-py35/532dc22/ : SUCCESS in 16m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/check/build-openstack-sphinx-docs/e9e3184/html/ : SUCCESS in 5m 57s\n- tempest-full http://logs.openstack.org/37/549737/3/check/tempest-full/1272f84/ : SUCCESS in 1h 25m 53s\n- neutron-grenade http://logs.openstack.org/37/549737/3/check/neutron-grenade/b25a439/ : SUCCESS in 45m 23s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/check/tempest-full-py3/2daa3fc/ : SUCCESS in 1h 22m 05s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/3/check/neutron-grenade-multinode/b4d1b0f/ : SUCCESS in 57m 09s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/c7e3887/ : SUCCESS in 53m 48s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-cells/4c51936/ : SUCCESS in 37m 57s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/83c3858/ : SUCCESS in 59m 51s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/3/check/neutron-tempest-multinode-full/8f099b0/ : FAILURE in 1h 19m 32s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-multinode-live-migration/1a4a32d/ : SUCCESS in 38m 32s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/3/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/387d88a/ : SUCCESS in 39m 13s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/3/check/nova-multiattach/241a4ce/ : SUCCESS in 54m 02s\n- nova-tox-functional http://logs.openstack.org/37/549737/3/check/nova-tox-functional/abff783/ : SUCCESS in 19m 45s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/check/nova-tox-functional-py35/f29d1b7/ : SUCCESS in 26m 24s\n- nova-next http://logs.openstack.org/37/549737/3/check/nova-next/223b9e5/ : SUCCESS in 1h 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4e77bf6b0fa49c56921b5031f8b2ae1f4fa843e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-06 21:56:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"304684464fb7b1d0725fbe0b8e1294b0ccfcb5c8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-03-06 22:21:08.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/549737/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/710c912 : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":3},{"id":"3c1bedf9a4cdb8a901aa02ae943d21e89faa886f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-03-06 22:25:23.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-549737-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"51f4563689ac7bd08a89ef54cd4e5dfa285a8d93","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-06 22:48:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/37/549737/3/check-vote/ext-nova-zuul/c49e50d : FAILURE in 51m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cd63ca88139e16c327fb17a1c7c43ae2b629f5e1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-03-06 22:58:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/549737/3/ : SUCCESS in 2h 41m 34s","accounts_in_message":[],"_revision_number":3},{"id":"6b0dea3b8034696077153ee3636f861b93c134c1","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-03-06 22:59:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/37/549737/3/check/nova-out-of-tree-pvm/de4589b : SUCCESS in 1h 02m 48s\n- nova-in-tree-pvm http://184.172.12.213/37/549737/3/check/nova-in-tree-pvm/f7e70ec : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":3},{"id":"dd40b1e4697998cc228c7a400c26cb2f2a06ca9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 00:07:38.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/gate/openstack-tox-pep8/3b138cd/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py27/9bb8db9/ : SUCCESS in 11m 14s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py35/f1db1f3/ : SUCCESS in 12m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/gate/build-openstack-sphinx-docs/99f6c83/html/ : SUCCESS in 5m 53s\n- tempest-full tempest-full : MERGER_FAILURE\n- neutron-grenade http://logs.openstack.org/37/549737/3/gate/neutron-grenade/d6d0c0e/ : SUCCESS in 45m 07s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/gate/tempest-full-py3/45297fe/ : SUCCESS in 1h 27m 14s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-cells/72490e7/ : SUCCESS in 38m 59s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-multinode-live-migration/6195886/ : SUCCESS in 35m 19s\n- nova-multiattach nova-multiattach : MERGER_FAILURE\n- nova-tox-functional http://logs.openstack.org/37/549737/3/gate/nova-tox-functional/3bf2d23/ : SUCCESS in 22m 44s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/gate/nova-tox-functional-py35/bd9e8ec/ : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2b041220d549c9c6381eb5d0a97b733122cb4ffd","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-03-07 08:02:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"be08b0b23b1775df26c9c69a05f3571af6feb1ed","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-07 08:55:36.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14836 : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5f178e9766c9b236d6653711d5dfff0ca66819b0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-03-07 09:00:46.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/549737/3/ : FAILURE in 57m 20s","accounts_in_message":[],"_revision_number":3},{"id":"099e6069125a32054f76acc8e563de0cf1dd2326","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-03-07 09:00:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/549737/3/check/check-dsvm-tempest-vz7-exe-minimal/7cd8e48 : SUCCESS in 57m 11s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"a17928441ec286030c7d28f2144d0a6658a3bd9c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-07 09:04:21.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/tempest-dsvm-full-xenial/785966e/ : SUCCESS in 1h 00m 45s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/grenade-dsvm-xenial/78e1435/ : SUCCESS in 40m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1da9d7026355bd1753246256ad34441bf57402f5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-07 09:07:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-intel-nfv-xenial/bf1d7f1 : SUCCESS in 27m 06s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/4f86855 : SUCCESS in 1h 03m 47s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/0ad1ee4 : SUCCESS in 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6688d4df880edf4559d8343583fd72987a753ce7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-07 09:08:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/37/549737/3/check-vote/ext-nova-zuul/ccb3d9b : FAILURE in 1h 05m 03s","accounts_in_message":[],"_revision_number":3},{"id":"79825577f185da3a9ee9b863f18d17e2a4336c46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 09:46:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/check/openstack-tox-pep8/80b1b02/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/check/openstack-tox-py27/da6c7c5/ : SUCCESS in 10m 35s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/check/openstack-tox-py35/d0ebde7/ : SUCCESS in 12m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/check/build-openstack-sphinx-docs/54d19ed/html/ : SUCCESS in 6m 00s\n- tempest-full http://logs.openstack.org/37/549737/3/check/tempest-full/057dffd/ : SUCCESS in 1h 37m 38s\n- neutron-grenade http://logs.openstack.org/37/549737/3/check/neutron-grenade/53c8a76/ : SUCCESS in 44m 23s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/check/tempest-full-py3/1a76115/ : SUCCESS in 1h 10m 45s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/3/check/neutron-grenade-multinode/2429af6/ : SUCCESS in 54m 04s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/8d3dbb8/ : SUCCESS in 1h 00m 29s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-cells/da75705/ : SUCCESS in 48m 27s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/585916a/ : SUCCESS in 1h 16m 12s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/3/check/neutron-tempest-multinode-full/48c4390/ : SUCCESS in 1h 31m 22s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-multinode-live-migration/5abc742/ : SUCCESS in 38m 18s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/3/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9ef84c5/ : SUCCESS in 1h 00m 50s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/3/check/nova-multiattach/a84c926/ : SUCCESS in 54m 50s\n- nova-tox-functional http://logs.openstack.org/37/549737/3/check/nova-tox-functional/71165ef/ : SUCCESS in 20m 51s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/check/nova-tox-functional-py35/f8cfcaf/ : SUCCESS in 20m 26s\n- nova-next http://logs.openstack.org/37/549737/3/check/nova-next/c00ab6c/ : SUCCESS in 1h 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dd36382f29a22db1a3af0cd0f9f1c3c681c98adf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 09:46:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ec394094f8f69013984adec0d135d3ab75980d9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-03-07 10:35:25.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- dsvm-nova http://207.189.188.190/logs/37/549737/3/check-vote/ext-nova-zuul/bbf34e4 : FAILURE in 48m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3dce30f7c1fb67fb8d2cf8c4b3569f800a6299dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 10:40:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/gate/openstack-tox-pep8/33c31d6/ : RETRY_LIMIT in 36s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py27/5b71836/ : RETRY_LIMIT in 33s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py35/439a1f5/ : RETRY_LIMIT in 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/gate/build-openstack-sphinx-docs/f70f2fb/ : RETRY_LIMIT in 30s\n- tempest-full http://logs.openstack.org/37/549737/3/gate/tempest-full/3ce9215/ : RETRY_LIMIT in 4m 44s\n- neutron-grenade http://logs.openstack.org/37/549737/3/gate/neutron-grenade/e8a6e9c/ : RETRY_LIMIT in 33s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/gate/tempest-full-py3/98aa60c/ : RETRY_LIMIT in 4m 48s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-cells/19ad37a/ : RETRY_LIMIT in 29s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-multinode-live-migration/7a2048c/ : RETRY_LIMIT in 29s\n- nova-multiattach http://logs.openstack.org/37/549737/3/gate/nova-multiattach/d18b23e/ : RETRY_LIMIT in 35s\n- nova-tox-functional http://logs.openstack.org/37/549737/3/gate/nova-tox-functional/5a015cf/ : RETRY_LIMIT in 37s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/gate/nova-tox-functional-py35/8aca220/ : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":3},{"id":"c75f9b0aed869d7a602e94eb763b7499725f7c42","author":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"date":"2018-03-07 13:09:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"792381cf7cbce04fd2c23d8a3117c318e7469296","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-03-07 14:02:29.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-14851 : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":3},{"id":"44d7f09e10e8d40aec1cd5765289d8467facd82c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-03-07 14:11:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/37/549737/3/check/check-dsvm-tempest-vz7-exe-minimal/447eba8 : SUCCESS in 1h 00m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"99239af0530d4e0ceb915a9a35646dad0ae80933","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-03-07 14:18:25.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/tempest-dsvm-full-xenial/159fb76/ : SUCCESS in 1h 07m 51s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/37/549737/3/check/grenade-dsvm-xenial/dd4de7e/ : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"04dd69b193018220e79794f371b2d9c94d9d0925","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 15:01:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/check/openstack-tox-pep8/189d61d/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/check/openstack-tox-py27/e1c2573/ : SUCCESS in 12m 38s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/check/openstack-tox-py35/ca02e74/ : SUCCESS in 12m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/check/build-openstack-sphinx-docs/5f3796b/html/ : SUCCESS in 5m 47s\n- tempest-full http://logs.openstack.org/37/549737/3/check/tempest-full/cd82b2c/ : SUCCESS in 1h 18m 37s\n- neutron-grenade http://logs.openstack.org/37/549737/3/check/neutron-grenade/00887a9/ : SUCCESS in 57m 08s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/check/tempest-full-py3/8f97dd1/ : SUCCESS in 1h 06m 58s\n- neutron-grenade-multinode http://logs.openstack.org/37/549737/3/check/neutron-grenade-multinode/231ba04/ : SUCCESS in 49m 57s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/f3a8fbc/ : SUCCESS in 1h 02m 43s (non-voting)\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-cells/9546fba/ : SUCCESS in 49m 11s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/8119e6b/ : SUCCESS in 1h 02m 28s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/37/549737/3/check/neutron-tempest-multinode-full/a3cd84f/ : SUCCESS in 1h 33m 38s (non-voting)\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/check/legacy-tempest-dsvm-multinode-live-migration/439557d/ : SUCCESS in 43m 00s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/37/549737/3/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/acfe006/ : SUCCESS in 40m 38s (non-voting)\n- nova-multiattach http://logs.openstack.org/37/549737/3/check/nova-multiattach/a694ab0/ : SUCCESS in 1h 20m 20s\n- nova-next http://logs.openstack.org/37/549737/3/check/nova-next/cfae78d/ : SUCCESS in 1h 14m 20s (non-voting)\n- nova-tox-functional http://logs.openstack.org/37/549737/3/check/nova-tox-functional/f0687a7/ : SUCCESS in 19m 13s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/check/nova-tox-functional-py35/c08f4e1/ : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2658eb685b1b0ad8f265266aab19dcc31e042efd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 15:02:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8de52812c8acf781cfe52a43e13a8b9b644221d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-03-07 15:54:42.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-37-549737-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"36bcd61b7ad043f23ea75adac6ae6b695abad0b0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-03-07 16:53:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-intel-nfv-xenial/59b85b5 : SUCCESS in 29m 52s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/0756082 : SUCCESS in 1h 02m 33s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-03-07/549737/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6fddcc5 : SUCCESS in 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c2b60c3bf6168ac74a65a4aca1fa4ff7bca73a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 17:26:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4a9ff9535b4e5a1c7bd5f60b9ffaf1dfc0a75713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 17:26:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/549737/3/gate/openstack-tox-pep8/93839e4/ : SUCCESS in 8m 16s\n- openstack-tox-py27 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py27/903c7e8/ : SUCCESS in 11m 12s\n- openstack-tox-py35 http://logs.openstack.org/37/549737/3/gate/openstack-tox-py35/b2c0ebb/ : SUCCESS in 12m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/37/549737/3/gate/build-openstack-sphinx-docs/e534370/html/ : SUCCESS in 6m 11s\n- tempest-full http://logs.openstack.org/37/549737/3/gate/tempest-full/64e7497/ : SUCCESS in 1h 24m 42s\n- neutron-grenade http://logs.openstack.org/37/549737/3/gate/neutron-grenade/f7e9e42/ : SUCCESS in 57m 16s\n- tempest-full-py3 http://logs.openstack.org/37/549737/3/gate/tempest-full-py3/079dc3a/ : SUCCESS in 1h 16m 59s\n- legacy-tempest-dsvm-cells http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-cells/36ad5e6/ : SUCCESS in 39m 19s\n- legacy-tempest-dsvm-multinode-live-migration http://logs.openstack.org/37/549737/3/gate/legacy-tempest-dsvm-multinode-live-migration/3f34190/ : SUCCESS in 36m 24s\n- nova-multiattach http://logs.openstack.org/37/549737/3/gate/nova-multiattach/2e3cf37/ : SUCCESS in 47m 57s\n- nova-tox-functional http://logs.openstack.org/37/549737/3/gate/nova-tox-functional/16f0b6a/ : SUCCESS in 22m 43s\n- nova-tox-functional-py35 http://logs.openstack.org/37/549737/3/gate/nova-tox-functional-py35/86051a4/ : SUCCESS in 21m 15s","accounts_in_message":[],"_revision_number":3},{"id":"146adcf6b9e3c52313e70843a4a5598a401c2544","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-03-07 18:25:29.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 0ba8721e5451db0e047185b0b49d780727c40a2b","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b81a66b3b0b1ce9f56280dda40ac871132cb29f2","revisions":{"46718e70704ca97aed8c79055cf1615b7d78ff25":{"kind":"REWORK","_number":1,"created":"2018-03-05 10:59:08.000000000","uploader":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"ref":"refs/changes/37/549737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/37/549737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/37/549737/1"}}},"commit":{"parents":[{"commit":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","subject":"Merge \"Move makefs to privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6edacd66c6fca1ab4f132253462f4c15eb9b217"}]}],"author":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-05 10:48:28.000000000","tz":0},"committer":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-05 10:57:20.000000000","tz":0},"subject":"Fix version cap when no nova-compute started","message":"Fix version cap when no nova-compute started\n\nThe version cap was set to mitaka (4.11) for newton and has not been\nchanged since. Upgrades are only supported for version N to N+1. The bug\nhas become more apparent with the removal of compute RPC 4.x\ncompatibility (a761e57368280b6d3e931831ecd393fd5787b3ef) and can be seen\nwhen upgrade_levels/compute\u003dauto is set on a new deployment.\n\nThis change updates the default to queens to prevent the use of RPC 4.x\nin requests.\n\nChange-Id: I696a8ea8adbe9481e11407ecafd5e47b2bd29804\nPartial-bug: 1753443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46718e70704ca97aed8c79055cf1615b7d78ff25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46718e70704ca97aed8c79055cf1615b7d78ff25"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c8c49d6baa2b63ac87586410f21d05c21828e57a":{"kind":"REWORK","_number":2,"created":"2018-03-06 17:38:10.000000000","uploader":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"ref":"refs/changes/37/549737/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/37/549737/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/37/549737/2"}}},"commit":{"parents":[{"commit":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","subject":"Merge \"Move makefs to privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6edacd66c6fca1ab4f132253462f4c15eb9b217"}]}],"author":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-05 10:48:28.000000000","tz":0},"committer":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-06 17:37:26.000000000","tz":0},"subject":"Fix version cap when no nova-compute started","message":"Fix version cap when no nova-compute started\n\nWhen a zero service version is returned, it means that we have no\nservices running for the requested binary. In that case, we should\nassume the latest version available until told otherwise. This usually\nhappens in first-start cases, where everything is likely to be up to\ndate anyway.\n\nThis change addresses an issue where the version returned had been\nhard-coded to 4.11 (mitaka).\n\nChange-Id: I696a8ea8adbe9481e11407ecafd5e47b2bd29804\nCloses-bug: 1753443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c8c49d6baa2b63ac87586410f21d05c21828e57a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c8c49d6baa2b63ac87586410f21d05c21828e57a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b81a66b3b0b1ce9f56280dda40ac871132cb29f2":{"kind":"REWORK","_number":3,"created":"2018-03-06 20:15:56.000000000","uploader":{"_account_id":7219,"name":"git-harry","email":"git-harry@live.co.uk","username":"git-harry"},"ref":"refs/changes/37/549737/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/37/549737/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/37/549737/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/37/549737/3"}}},"commit":{"parents":[{"commit":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","subject":"Merge \"Move makefs to privsep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6edacd66c6fca1ab4f132253462f4c15eb9b217"}]}],"author":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-05 10:48:28.000000000","tz":0},"committer":{"name":"git-harry","email":"git-harry@live.co.uk","date":"2018-03-06 20:14:42.000000000","tz":0},"subject":"Fix version cap when no nova-compute started","message":"Fix version cap when no nova-compute started\n\nWhen a zero service version is returned, it means that we have no\nservices running for the requested binary. In that case, we should\nassume the latest version available until told otherwise. This usually\nhappens in first-start cases, where everything is likely to be up to\ndate anyway.\n\nThis change addresses an issue where the version returned had been\nhard-coded to 4.11 (mitaka).\n\nChange-Id: I696a8ea8adbe9481e11407ecafd5e47b2bd29804\nCloses-bug: 1753443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b81a66b3b0b1ce9f56280dda40ac871132cb29f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b81a66b3b0b1ce9f56280dda40ac871132cb29f2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a6edacd66c6fca1ab4f132253462f4c15eb9b217","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
