)]}'
{"id":"openstack%2Fnova~56075","triplet_id":"openstack%2Fnova~master~I6346281ca3516c31b195ce71942786d724e5e9df","project":"openstack/nova","branch":"master","topic":"(detached","hashtags":[],"change_id":"I6346281ca3516c31b195ce71942786d724e5e9df","subject":"Cache Neutron Client for Admin Scenarios","status":"MERGED","created":"2013-11-12 18:28:33.000000000","updated":"2013-11-15 23:10:03.000000000","submitted":"2013-11-14 23:35:18.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"353b84d9ad628980366548811a3db5541c3db16b","_number":56075,"virtual_id_number":56075,"owner":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-11-14 23:35:18.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":1,"date":"2013-11-14 04:24:29.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"recommended":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":1,"date":"2013-11-14 05:29:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":2,"date":"2013-11-14 20:27:46.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":1,"date":"2013-11-14 05:38:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2013-11-14 04:27:05.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},{"value":1,"date":"2013-11-14 14:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":1,"date":"2013-11-14 20:27:46.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"date":"2013-11-14 04:27:05.000000000","_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"value":0,"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},{"value":0,"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-12 18:44:46.000000000","updated_by":{"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},"reviewer":{"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},"state":"REVIEWER"},{"updated":"2013-11-13 14:43:58.000000000","updated_by":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"reviewer":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"state":"REVIEWER"},{"updated":"2013-11-14 04:24:29.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-11-14 04:27:05.000000000","updated_by":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"reviewer":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"state":"REVIEWER"},{"updated":"2013-11-14 05:29:49.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2013-11-14 05:38:17.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2013-11-14 14:10:00.000000000","updated_by":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"reviewer":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"state":"REVIEWER"},{"updated":"2013-11-14 20:27:46.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"},{"updated":"2013-11-14 23:35:18.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7289ad5e6c90ef6e34480d578d9ed8762b12dffe","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 18:28:52.000000000","message":"Patch Set 1: Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d4dc7cc2534f6222dcea12070139b2386aa5e349","author":{"_account_id":6316,"name":"Anita Kuno","email":"anteaya@anteaya.info","username":"anteaya"},"date":"2013-11-12 18:44:46.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"e6a0686cb5ded255c871e2259a55055848b4f5e5","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-12 18:48:33.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156411 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156412 : SUCCESS \n","accounts_in_message":[],"_revision_number":1},{"id":"f7f5455a725854f6706d47ce7d759920a9f6c1e7","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-12 18:50:27.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nI would like to see neutronclient itself handle caching of tokens not nova.","accounts_in_message":[],"_revision_number":1},{"id":"de6f434d28cd9b613fcdb4d76bfa4cc80604fd79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 19:07:07.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/1/check/gate-nova-pep8/310a7a6 : FAILURE in 5m 18s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/1/check/gate-nova-docs/946f1f1/doc/build/html/ : SUCCESS in 4m 19s\n- gate-nova-python26 http://logs.openstack.org/75/56075/1/check/gate-nova-python26/df377da : SUCCESS in 21m 42s\n- gate-nova-python27 http://logs.openstack.org/75/56075/1/check/gate-nova-python27/aaef94f : SUCCESS in 16m 21s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/1/check/check-tempest-devstack-vm-full/fbc6098 : SUCCESS in 34m 25s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/1/check/check-tempest-devstack-vm-postgres-full/75c8e99 : SUCCESS in 30m 57s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/1/check/check-tempest-devstack-vm-neutron/e811f99 : FAILURE in 14m 49s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/1/check/gate-tempest-devstack-vm-large-ops/72ba8b3 : SUCCESS in 14m 45s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/1/check/gate-tempest-devstack-vm-neutron-large-ops/6f4787d : FAILURE in 38m 12s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/1/check/check-grenade-devstack-vm/033d0d7 : SUCCESS in 37m 47s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/1/check/gate-devstack-vm-cells/ecf2a67 : SUCCESS in 8m 56s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/1/check/gate-nova-pylint/0354bc8 : SUCCESS in 14m 02s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"1c0ad15b00685078e20c0c7daa23f040746b160d","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-12 19:15:17.000000000","message":"Patch Set 1:\n\nto clarify, I am fine with this an interim solution while the keystone folks work and making all clients properly cache tokens.  I would like to see a comment in the code saying this logic can go away when BP x is completed.","accounts_in_message":[],"_revision_number":1},{"id":"4f478dc4ffc46d117ab28425ccc00980a739bd64","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 19:19:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e8851409d637a014b49f0a84dab210b58c22c02","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 19:22:21.000000000","message":"Patch Set 2: Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"4aa2d6ee2212ed442bebf5ce10c27c97894fe78a","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 19:23:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c8146ca7ff88a00d34c8b555d1b6fc164843e689","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-11-12 19:24:13.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":3},{"id":"f6c0876aed44834b52db3d26184e307f770f5531","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-12 19:46:03.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156449 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156450 : SUCCESS \n","accounts_in_message":[],"_revision_number":3},{"id":"d707daba442639590b6de8f5acf36d2c07c7b582","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 20:07:39.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/3/check/gate-nova-pep8/b497320 : SUCCESS in 4m 42s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/3/check/gate-nova-docs/bc7949c/doc/build/html/ : SUCCESS in 4m 12s\n- gate-nova-python26 http://logs.openstack.org/75/56075/3/check/gate-nova-python26/eac2754 : SUCCESS in 21m 16s\n- gate-nova-python27 http://logs.openstack.org/75/56075/3/check/gate-nova-python27/d8dcfd9 : SUCCESS in 15m 40s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/3/check/check-tempest-devstack-vm-full/8c012b4 : SUCCESS in 32m 52s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/3/check/check-tempest-devstack-vm-postgres-full/064bb1d : SUCCESS in 31m 24s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/3/check/check-tempest-devstack-vm-neutron/2646886 : FAILURE in 14m 50s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/3/check/gate-tempest-devstack-vm-large-ops/bf1389b : SUCCESS in 17m 38s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/3/check/gate-tempest-devstack-vm-neutron-large-ops/6af4c51 : FAILURE in 37m 32s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/3/check/check-grenade-devstack-vm/e11d616 : SUCCESS in 43m 23s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/3/check/gate-devstack-vm-cells/2da7be6 : SUCCESS in 8m 48s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/3/check/gate-nova-pylint/12e041d : SUCCESS in 10m 12s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"4672e46d802bbe82182f0d33fa3660fce50ca1b2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 20:56:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"696907f11ff6f524864899aba31c3a22623ee52f","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 20:57:20.000000000","message":"Patch Set 4: Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"0906b0f274185d87390e2eb9793377fb4d0d8d0b","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-12 21:21:17.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156475 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156476 : SUCCESS \n","accounts_in_message":[],"_revision_number":4},{"id":"2c1d40763a327b3c3c24455591f48939560fd53a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 21:36:17.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/4/check/gate-nova-pep8/95ab113 : SUCCESS in 3m 18s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/4/check/gate-nova-docs/3edce57/doc/build/html/ : SUCCESS in 4m 00s\n- gate-nova-python26 http://logs.openstack.org/75/56075/4/check/gate-nova-python26/ed10e4e : SUCCESS in 22m 51s\n- gate-nova-python27 http://logs.openstack.org/75/56075/4/check/gate-nova-python27/977b223 : SUCCESS in 13m 26s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/4/check/check-tempest-devstack-vm-full/ab84e15 : SUCCESS in 31m 20s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/4/check/check-tempest-devstack-vm-postgres-full/ce369e1 : SUCCESS in 31m 37s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/4/check/check-tempest-devstack-vm-neutron/123d6ae : FAILURE in 18m 00s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/4/check/gate-tempest-devstack-vm-large-ops/f5927c5 : SUCCESS in 14m 06s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/4/check/gate-tempest-devstack-vm-neutron-large-ops/2df02aa : FAILURE in 38m 19s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/4/check/check-grenade-devstack-vm/d17b775 : SUCCESS in 31m 58s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/4/check/gate-devstack-vm-cells/01cef76 : SUCCESS in 10m 40s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/4/check/gate-nova-pylint/42aa7bf : SUCCESS in 14m 54s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"b87b02cc07078c5640d12e981c172c01fdf45e3d","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 21:49:41.000000000","message":"Patch Set 4:\n\nrecheck bug 1232303","accounts_in_message":[],"_revision_number":4},{"id":"3cfcb695eb5b838086e972302dc9abb0a276a01b","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 22:15:07.000000000","message":"Patch Set 4: Rebased","accounts_in_message":[],"_revision_number":4},{"id":"e12c058cbe15c0c00caffd8b37c9edfc9375818f","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-12 22:41:05.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156493 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156494 : SUCCESS \n","accounts_in_message":[],"_revision_number":5},{"id":"615b3d35109b0f09f917420e533b01b3c2cc7a00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 22:56:04.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/5/check/gate-nova-pep8/7facd7f : SUCCESS in 4m 42s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/5/check/gate-nova-docs/7779912/doc/build/html/ : SUCCESS in 4m 10s\n- gate-nova-python26 http://logs.openstack.org/75/56075/5/check/gate-nova-python26/90c665e : SUCCESS in 24m 44s\n- gate-nova-python27 http://logs.openstack.org/75/56075/5/check/gate-nova-python27/a6c99c2 : SUCCESS in 15m 49s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-full/5236f1f : SUCCESS in 29m 07s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-postgres-full/1bfff01 : SUCCESS in 38m 59s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-neutron/528e3a2 : FAILURE in 14m 55s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/5/check/gate-tempest-devstack-vm-large-ops/1392b78 : SUCCESS in 13m 38s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/5/check/gate-tempest-devstack-vm-neutron-large-ops/271bf45 : FAILURE in 40m 38s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/5/check/check-grenade-devstack-vm/d96dd24 : SUCCESS in 33m 39s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/5/check/gate-devstack-vm-cells/ba2c095 : SUCCESS in 8m 52s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/5/check/gate-nova-pylint/0356c35 : SUCCESS in 10m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"41eba6267655cf3e39c0c8bff78b80a9810492f4","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-12 23:18:01.000000000","message":"Patch Set 5:\n\nrecheck bug 1232303","accounts_in_message":[],"_revision_number":5},{"id":"2c91643e97132c0a866aafc95c9d6e742c5a1564","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-12 23:59:14.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/5/check/gate-nova-pep8/54f7856 : SUCCESS in 3m 18s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/5/check/gate-nova-docs/fe324b7/doc/build/html/ : SUCCESS in 4m 00s\n- gate-nova-python26 http://logs.openstack.org/75/56075/5/check/gate-nova-python26/217fe41 : SUCCESS in 21m 27s\n- gate-nova-python27 http://logs.openstack.org/75/56075/5/check/gate-nova-python27/b9166bc : SUCCESS in 16m 20s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-full/752e42e : SUCCESS in 29m 48s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-postgres-full/6c6dafb : SUCCESS in 31m 42s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/5/check/check-tempest-devstack-vm-neutron/72e0824 : FAILURE in 14m 55s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/5/check/gate-tempest-devstack-vm-large-ops/619fdb5 : SUCCESS in 13m 50s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/5/check/gate-tempest-devstack-vm-neutron-large-ops/b62c44f : FAILURE in 40m 58s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/5/check/check-grenade-devstack-vm/4c56149 : SUCCESS in 30m 38s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/5/check/gate-devstack-vm-cells/24eb293 : SUCCESS in 10m 44s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/5/check/gate-nova-pylint/4da1551 : SUCCESS in 14m 31s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"0a4abc347fde2f7818463372832343e7a6811cd9","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 01:03:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a2dafece49fb5886b5e42ef91ae6dbd5e16ecf0","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 01:06:52.000000000","message":"Patch Set 6: Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"4209afafc3ee093d1e5df2c999106090026780b1","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 01:26:02.000000000","message":"Patch Set 6: Works for me\n\nSmokeStack Results (patch set 6):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156543 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156544 : SUCCESS \n","accounts_in_message":[],"_revision_number":6},{"id":"c8b3f37242dd017ef2302f2a2b095cca3be437aa","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 01:33:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b68df39bb41f2f70181511e314491dd7edae48fb","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 01:34:18.000000000","message":"Patch Set 7: Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"b8a1c515c654551a720da151a5013e0e2124ab40","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 01:57:04.000000000","message":"Patch Set 7: Works for me\n\nSmokeStack Results (patch set 7):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156547 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156548 : SUCCESS \n","accounts_in_message":[],"_revision_number":7},{"id":"e7a29b55934f3f7417d20ebcfdf19c88dd06103d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 02:15:52.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/7/check/gate-nova-pep8/f7923c5 : SUCCESS in 4m 45s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/7/check/gate-nova-docs/d00b374/doc/build/html/ : SUCCESS in 4m 16s\n- gate-nova-python26 http://logs.openstack.org/75/56075/7/check/gate-nova-python26/9c88e47 : SUCCESS in 21m 23s\n- gate-nova-python27 http://logs.openstack.org/75/56075/7/check/gate-nova-python27/6fa389e : SUCCESS in 17m 43s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-full/c5a6631 : SUCCESS in 33m 27s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-postgres-full/09d5dbf : SUCCESS in 41m 35s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-neutron/855f37f : SUCCESS in 34m 01s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/7/check/gate-tempest-devstack-vm-large-ops/2bb0501 : SUCCESS in 13m 48s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/7/check/gate-tempest-devstack-vm-neutron-large-ops/b6a8002 : SUCCESS in 15m 55s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/7/check/check-grenade-devstack-vm/66113e1 : SUCCESS in 33m 29s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/7/check/gate-devstack-vm-cells/046a6c8 : SUCCESS in 10m 29s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/7/check/gate-nova-pylint/54db7de : SUCCESS in 13m 29s (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"ec84114130e195526f12a8036b4723b8e98fd69e","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 02:19:59.000000000","message":"Patch Set 7:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":7},{"id":"3ca21630e367a11c407d3ed66aca71df26580521","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 03:24:37.000000000","message":"Patch Set 7: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/7/check/gate-nova-pep8/727d121 : SUCCESS in 4m 49s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/7/check/gate-nova-docs/13c3bc3/doc/build/html/ : SUCCESS in 4m 19s\n- gate-nova-python26 http://logs.openstack.org/75/56075/7/check/gate-nova-python26/8d21b3a : SUCCESS in 24m 26s\n- gate-nova-python27 http://logs.openstack.org/75/56075/7/check/gate-nova-python27/c19dd75 : SUCCESS in 15m 47s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-full/6024cdf : SUCCESS in 1h 04m 23s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-postgres-full/5ccc23c : SUCCESS in 39m 03s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/7/check/check-tempest-devstack-vm-neutron/53494c6 : SUCCESS in 41m 35s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/7/check/gate-tempest-devstack-vm-large-ops/56eb070 : SUCCESS in 17m 42s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/7/check/gate-tempest-devstack-vm-neutron-large-ops/9258c6e : SUCCESS in 17m 19s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/7/check/check-grenade-devstack-vm/c4a5934 : FAILURE in 1h 00m 15s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/7/check/gate-devstack-vm-cells/4113b04 : SUCCESS in 9m 20s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/7/check/gate-nova-pylint/4cf6ecb : SUCCESS in 13m 53s (non-voting)\n","accounts_in_message":[],"_revision_number":7},{"id":"655ddf4a67a7537d6e6fc9bdf5b1e9d48ef10f1c","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 03:25:07.000000000","message":"Patch Set 7:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":7},{"id":"e51361fa51f153efe6b252f0f71eb4876aa568a2","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2013-11-13 03:25:34.000000000","message":"Patch Set 7: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nThe code looks good, just a little care in the comment.","accounts_in_message":[],"_revision_number":7},{"id":"3682ddb094d46e009247e0502ae96d36e535f26e","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 03:25:53.000000000","message":"Patch Set 7:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":7},{"id":"2117628128bece58c19ca268ad641fa483b7ecc3","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 03:30:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b91c051ed33dddf87a99216bb85a4ba4a9b4be68","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 03:31:09.000000000","message":"Patch Set 8: Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"26d12d1e6b80460b890fc80e22fe4e726a684054","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 03:52:40.000000000","message":"Patch Set 8: Works for me\n\nSmokeStack Results (patch set 8):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156576 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156577 : SUCCESS \n","accounts_in_message":[],"_revision_number":8},{"id":"082370829bea443a6ee97e97bcd038b3bbd694ef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 04:22:36.000000000","message":"Patch Set 8: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/8/check/gate-nova-pep8/ca859a4 : SUCCESS in 3m 13s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/8/check/gate-nova-docs/cb8f102/doc/build/html/ : SUCCESS in 4m 02s\n- gate-nova-python26 http://logs.openstack.org/75/56075/8/check/gate-nova-python26/3104c91 : SUCCESS in 23m 17s\n- gate-nova-python27 http://logs.openstack.org/75/56075/8/check/gate-nova-python27/bfc7bb3 : SUCCESS in 13m 50s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-full/46322a6 : SUCCESS in 31m 40s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-postgres-full/8d2a373 : SUCCESS in 31m 59s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-neutron/9d66c88 : SUCCESS in 51m 06s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/8/check/gate-tempest-devstack-vm-large-ops/3976544 : SUCCESS in 17m 36s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/8/check/gate-tempest-devstack-vm-neutron-large-ops/d951ce7 : SUCCESS in 15m 34s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/8/check/check-grenade-devstack-vm/06a4fe8 : SUCCESS in 38m 08s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/8/check/gate-devstack-vm-cells/a7b94b5 : SUCCESS in 8m 41s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/8/check/gate-nova-pylint/859368f : SUCCESS in 13m 44s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"9789e84757f0889f751367027317da7a04e9f0d9","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2013-11-13 09:46:07.000000000","message":"Patch Set 8: I would prefer that you didn\u0027t merge this\n\nI\u0027m not convinced of the logic here.   From what I can see this will cache an admin token for the life of thread, but threads in nova-compute are long lived and reused to serve multiple requests.  So it looks as if each thread in the rpc event pool (and the periodic task) would get its own admin token - but what happens when those tokens expire ?","accounts_in_message":[],"_revision_number":8},{"id":"113815313039b461ba678a27d0d13a5364ce4eb4","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 11:26:22.000000000","message":"Patch Set 8:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":8},{"id":"ad452d17cd986c493d375abdb5798b264340f1a2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 11:28:51.000000000","message":"Patch Set 8:\n\n@Phil, great question. In the neutron client, when the tokens expire (for admin scenario), there is enough information to generate a new one. So the neutron client just goes ahead and mints a new one and retries the operation. So we should be good. This was exactly the reasoning behind Drew\u0027s previous change to setup the credentials properly for the admin scenario so tokens can be re-genned.","accounts_in_message":[],"_revision_number":8},{"id":"ca22a680b5cfae95e79c59553ef324f5efdf123b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 12:11:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/8/check/gate-nova-pep8/d5ffb7d : SUCCESS in 5m 07s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/8/check/gate-nova-docs/a6cb831/doc/build/html/ : SUCCESS in 3m 00s\n- gate-nova-python26 http://logs.openstack.org/75/56075/8/check/gate-nova-python26/f3e5ec3 : SUCCESS in 25m 33s\n- gate-nova-python27 http://logs.openstack.org/75/56075/8/check/gate-nova-python27/d901d85 : SUCCESS in 17m 12s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-full/88b74f4 : SUCCESS in 29m 56s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-postgres-full/8d9403f : SUCCESS in 29m 42s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/8/check/check-tempest-devstack-vm-neutron/b5f192e : SUCCESS in 27m 11s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/8/check/gate-tempest-devstack-vm-large-ops/bb6b623 : SUCCESS in 14m 35s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/8/check/gate-tempest-devstack-vm-neutron-large-ops/fff5e7d : SUCCESS in 17m 24s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/8/check/check-grenade-devstack-vm/54a27b3 : SUCCESS in 44m 35s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/8/check/gate-devstack-vm-cells/f37408e : SUCCESS in 9m 09s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/8/check/gate-nova-pylint/24dc22b : SUCCESS in 13m 40s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"7540039c1ccd67fee25ff3d1e55108cd69f521d6","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 12:40:12.000000000","message":"Patch Set 8:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":8},{"id":"a7090792a1e58fa5177fbbdb69b2324514a52422","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 12:41:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cc90a9ddc8c83daa380a618e952dd50243d83f97","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 13:10:36.000000000","message":"Patch Set 9: Works for me\n\nSmokeStack Results (patch set 9):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156773 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156774 : SUCCESS \n","accounts_in_message":[],"_revision_number":9},{"id":"db2000d3b1a1c6152890f33793d5ab39837c1db4","author":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"date":"2013-11-13 13:22:07.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\n(1 inline comment)\n\nI think it is the right approach to cache the client vs the token in order to avoid dealing with token expiration.\nThread-local storage also seems reasonable.\n\nfwiw lgtm","accounts_in_message":[],"_revision_number":9},{"id":"69a3bf1d08a9f5f917afcafead4fb4c6fbb0d897","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 13:25:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1ba4c0f9a3248ec165693684dd03d3ac0d5a9be8","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 13:52:06.000000000","message":"Patch Set 10: Works for me\n\nSmokeStack Results (patch set 10):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156791 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156792 : SUCCESS \n","accounts_in_message":[],"_revision_number":10},{"id":"992c9fe574043a39d03969da2030108f62f4ba3e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 14:26:35.000000000","message":"Patch Set 10: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/10/check/gate-nova-pep8/108dbd1 : SUCCESS in 4m 51s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/10/check/gate-nova-docs/d63db1e/doc/build/html/ : SUCCESS in 4m 27s\n- gate-nova-python26 http://logs.openstack.org/75/56075/10/check/gate-nova-python26/dbe8ae4 : SUCCESS in 22m 49s\n- gate-nova-python27 http://logs.openstack.org/75/56075/10/check/gate-nova-python27/03b2e2c : SUCCESS in 16m 25s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-full/39b1e4c : SUCCESS in 30m 48s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-postgres-full/b88e2c3 : SUCCESS in 32m 04s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-neutron/1b505f6 : SUCCESS in 24m 41s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/10/check/gate-tempest-devstack-vm-large-ops/9824f7b : SUCCESS in 15m 09s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/10/check/gate-tempest-devstack-vm-neutron-large-ops/31bead5 : SUCCESS in 17m 38s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/10/check/check-grenade-devstack-vm/5a665cd : FAILURE in 1h 00m 13s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/10/check/gate-devstack-vm-cells/07aeca7 : SUCCESS in 8m 57s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/10/check/gate-nova-pylint/e9dc689 : SUCCESS in 15m 36s (non-voting)\n","accounts_in_message":[],"_revision_number":10},{"id":"c934b471d49903dc8582b2579700ff909650b24d","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2013-11-13 14:27:36.000000000","message":"Patch Set 10: Looks good to me, but someone else must approve\n\nThanks for the explanation on how tokens get refreshed.","accounts_in_message":[],"_revision_number":10},{"id":"44545b211a5d0d184d917607a2119922524b49d7","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 14:29:13.000000000","message":"Patch Set 10:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":10},{"id":"dd5aab6d782cc1298314ef136068169d245f8bcb","author":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"date":"2013-11-13 14:43:58.000000000","message":"Patch Set 10: Looks good to me, but someone else must approve\n\nStill fwiw, large ops now consistently pass in my environment as well, even if with bit of instrumentation I see all the VMs become active in average in 181 seconds, which still a bit close to the threshold of 196 seconds.\n\nSo this is definetely great, but we need more to claim stability on largeops.","accounts_in_message":[],"_revision_number":10},{"id":"95b0ebdfa89fc74bb52cf32b6a435bf179a9ce4a","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 14:46:38.000000000","message":"Patch Set 10:\n\n@Salvatore, definitely agree with your comments. For others, please follow details at https://bugs.launchpad.net/neutron/+bug/1250168","accounts_in_message":[],"_revision_number":10},{"id":"443391e1dde77a632b5d0bfa1729b16957f0aa93","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 15:14:28.000000000","message":"Patch Set 10: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/10/check/gate-nova-pep8/ae39400 : SUCCESS in 4m 46s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/10/check/gate-nova-docs/389c187/doc/build/html/ : SUCCESS in 3m 07s\n- gate-nova-python26 http://logs.openstack.org/75/56075/10/check/gate-nova-python26/57fb589 : SUCCESS in 18m 08s\n- gate-nova-python27 http://logs.openstack.org/75/56075/10/check/gate-nova-python27/5ff0268 : SUCCESS in 17m 07s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-full/6b8fce3 : SUCCESS in 44m 24s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-postgres-full/8991bf8 : SUCCESS in 34m 03s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/10/check/check-tempest-devstack-vm-neutron/a6a4831 : SUCCESS in 34m 46s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/10/check/gate-tempest-devstack-vm-large-ops/2f631eb : SUCCESS in 14m 27s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/10/check/gate-tempest-devstack-vm-neutron-large-ops/c96c89d : SUCCESS in 16m 03s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/10/check/check-grenade-devstack-vm/2288ae0 : SUCCESS in 44m 59s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/10/check/gate-devstack-vm-cells/e4f725b : SUCCESS in 8m 53s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/10/check/gate-nova-pylint/8477d30 : SUCCESS in 13m 04s (non-voting)\n","accounts_in_message":[],"_revision_number":10},{"id":"990f88e6f983fb2b3f20c1a6051aeda7130fe3d6","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-13 17:22:48.000000000","message":"Patch Set 10: Looks good to me, but someone else must approve\n\nLGTM, looks like this patch makes things significantly faster.  I would prefer to see some unit tests with this patch then I am +2.","accounts_in_message":[],"_revision_number":10},{"id":"44b4c940caba8065acd097b1746283b1d050f142","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-13 17:26:44.000000000","message":"Patch Set 10: (1 inline comment)\n\ninline comment","accounts_in_message":[],"_revision_number":10},{"id":"a0706be04c095cd86e0311516647507af65caf77","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 18:04:08.000000000","message":"Patch Set 10: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":10},{"id":"f1e3103b4bf320c3953098513c5b333bc486c216","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 19:53:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"23c7f081c56aca47c4b2a62deff70b27b2555694","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 19:55:04.000000000","message":"Patch Set 11: Rebased","accounts_in_message":[],"_revision_number":11},{"id":"ac065f936f4130f99dd9b18b6099bbcdfc35cd4a","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 20:17:25.000000000","message":"Patch Set 12: Works for me\n\nSmokeStack Results (patch set 12):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156904 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156905 : SUCCESS \n","accounts_in_message":[],"_revision_number":12},{"id":"a024b83438b4e6bd2aed23204a31b7f16bc9c2f2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 20:40:46.000000000","message":"Patch Set 12: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/12/check/gate-nova-pep8/348bfac : SUCCESS in 5m 31s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/12/check/gate-nova-docs/84bdef7/doc/build/html/ : SUCCESS in 3m 49s\n- gate-nova-python26 http://logs.openstack.org/75/56075/12/check/gate-nova-python26/154d41f : SUCCESS in 21m 33s\n- gate-nova-python27 http://logs.openstack.org/75/56075/12/check/gate-nova-python27/030477a : SUCCESS in 17m 26s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-full/fe3fb06 : SUCCESS in 33m 36s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-postgres-full/6671dbf : SUCCESS in 43m 05s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-neutron/bb70747 : SUCCESS in 26m 40s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/12/check/gate-tempest-devstack-vm-large-ops/cbc0994 : SUCCESS in 15m 13s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/12/check/gate-tempest-devstack-vm-neutron-large-ops/438afdb : FAILURE in 16m 06s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/12/check/check-grenade-devstack-vm/83a8394 : SUCCESS in 44m 58s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/12/check/gate-devstack-vm-cells/47db992 : SUCCESS in 9m 09s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/12/check/gate-nova-pylint/1b889bc : SUCCESS in 14m 04s (non-voting)\n","accounts_in_message":[],"_revision_number":12},{"id":"ccb522a6f629afe5a2e45b5b896538551e848608","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-13 20:41:20.000000000","message":"Patch Set 12:\n\nhmm we just failed large-ops again","accounts_in_message":[],"_revision_number":12},{"id":"65717a9b9ec90f4971c0198309314c6f06d3fee6","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-13 20:43:05.000000000","message":"Patch Set 12:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":12},{"id":"90280fa85007d69a56362bcb5821feb04bcf5e49","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 21:26:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/12/check/gate-nova-pep8/2253e27 : SUCCESS in 4m 03s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/12/check/gate-nova-docs/4c42587/doc/build/html/ : SUCCESS in 4m 22s\n- gate-nova-python26 http://logs.openstack.org/75/56075/12/check/gate-nova-python26/4159c85 : SUCCESS in 23m 14s\n- gate-nova-python27 http://logs.openstack.org/75/56075/12/check/gate-nova-python27/b0c6b2c : SUCCESS in 18m 26s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-full/2f66d04 : SUCCESS in 43m 00s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-postgres-full/88b1c48 : SUCCESS in 32m 09s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/12/check/check-tempest-devstack-vm-neutron/9498ae0 : SUCCESS in 29m 30s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/12/check/gate-tempest-devstack-vm-large-ops/012f293 : SUCCESS in 18m 15s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/12/check/gate-tempest-devstack-vm-neutron-large-ops/e66d250 : SUCCESS in 16m 02s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/12/check/check-grenade-devstack-vm/ec2e9e3 : SUCCESS in 32m 45s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/12/check/gate-devstack-vm-cells/ddec01e : SUCCESS in 9m 06s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/12/check/gate-nova-pylint/6e4f8b9 : SUCCESS in 13m 55s (non-voting)\n","accounts_in_message":[],"_revision_number":12},{"id":"affacf3d28f6648d874d8290d5505ba8dc421154","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 22:09:17.000000000","message":"Patch Set 12:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":12},{"id":"3b98d45a27da84913905611ce1c7e6528febc1c5","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 22:38:00.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ac626b4119115469037698f02e04be116a630e14","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2013-11-13 23:00:18.000000000","message":"Patch Set 13: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":13},{"id":"f45d324b2a04b8890725408a406c4628eeb42d7b","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-13 23:00:28.000000000","message":"Patch Set 13: Works for me\n\nSmokeStack Results (patch set 13):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156977 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156978 : SUCCESS \n","accounts_in_message":[],"_revision_number":13},{"id":"74f544091ed6468d7f78108e606374f0d6bbab50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-13 23:16:46.000000000","message":"Patch Set 13: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/13/check/gate-nova-pep8/cc3088b : FAILURE in 4m 38s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/13/check/gate-nova-docs/a59949b/doc/build/html/ : SUCCESS in 3m 04s\n- gate-nova-python26 http://logs.openstack.org/75/56075/13/check/gate-nova-python26/1380c52 : SUCCESS in 23m 17s\n- gate-nova-python27 http://logs.openstack.org/75/56075/13/check/gate-nova-python27/a416a52 : SUCCESS in 17m 36s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/13/check/check-tempest-devstack-vm-full/4613a7d : SUCCESS in 30m 35s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/13/check/check-tempest-devstack-vm-postgres-full/97688a2 : SUCCESS in 32m 27s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/13/check/check-tempest-devstack-vm-neutron/f1a2177 : SUCCESS in 25m 15s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/13/check/gate-tempest-devstack-vm-large-ops/6cdd39b : SUCCESS in 14m 01s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/13/check/gate-tempest-devstack-vm-neutron-large-ops/2ff85f4 : SUCCESS in 15m 50s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/13/check/check-grenade-devstack-vm/e766f06 : SUCCESS in 38m 25s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/13/check/gate-devstack-vm-cells/965c5bf : SUCCESS in 10m 49s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/13/check/gate-nova-pylint/139029f : SUCCESS in 14m 25s (non-voting)\n","accounts_in_message":[],"_revision_number":13},{"id":"e6f012342352309ec31c6e14e5507e88fbe2b362","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-13 23:50:45.000000000","message":"Patch Set 13: Rebased","accounts_in_message":[],"_revision_number":13},{"id":"458aa31e1cd49fe8c20c77ee0bdb3fec9e599c9c","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2013-11-13 23:51:04.000000000","message":"Patch Set 14: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":14},{"id":"dece09fb48efb3614bf1ccf4e4eb5d3d4bee996e","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2013-11-14 00:17:45.000000000","message":"Patch Set 14: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":14},{"id":"b652ce0ff195106b4a480e74fdf28d1239859a8b","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-14 00:18:01.000000000","message":"Patch Set 14: Works for me\n\nSmokeStack Results (patch set 14):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/156989 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/156990 : SUCCESS \n","accounts_in_message":[],"_revision_number":14},{"id":"ac799d3637e623cf082dbbc1f1bbab416f401181","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 01:04:26.000000000","message":"Patch Set 14: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/14/check/gate-nova-pep8/1604f3d : FAILURE in 5m 24s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/14/check/gate-nova-docs/5d29fb2/doc/build/html/ : SUCCESS in 4m 41s\n- gate-nova-python26 http://logs.openstack.org/75/56075/14/check/gate-nova-python26/4667746 : SUCCESS in 21m 23s\n- gate-nova-python27 http://logs.openstack.org/75/56075/14/check/gate-nova-python27/00a3be6 : SUCCESS in 16m 19s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/14/check/check-tempest-devstack-vm-full/3f5e525 : FAILURE in 1h 13m 10s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/14/check/check-tempest-devstack-vm-postgres-full/0c0d10f : SUCCESS in 30m 10s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/14/check/check-tempest-devstack-vm-neutron/7976f8b : SUCCESS in 25m 12s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/14/check/gate-tempest-devstack-vm-large-ops/341ef90 : SUCCESS in 14m 28s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/14/check/gate-tempest-devstack-vm-neutron-large-ops/46e6617 : SUCCESS in 14m 58s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/14/check/check-grenade-devstack-vm/05484a2 : SUCCESS in 32m 22s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/14/check/gate-devstack-vm-cells/33e5310 : SUCCESS in 9m 06s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/14/check/gate-nova-pylint/4053712 : SUCCESS in 13m 37s (non-voting)\n","accounts_in_message":[],"_revision_number":14},{"id":"0c8e1ebbb06867131d9a6a66bc94394bf13dca9b","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-14 01:10:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5b233a6c2a41da679e1e4adac67fc0366b4a0e48","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-14 01:33:06.000000000","message":"Patch Set 15: Works for me\n\nSmokeStack Results (patch set 15):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/157014 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/157015 : SUCCESS \n","accounts_in_message":[],"_revision_number":15},{"id":"75a14556fea7c904db8e00050a24346a8baf3dda","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 01:42:43.000000000","message":"Patch Set 15: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/15/check/gate-nova-pep8/390ac87 : SUCCESS in 4m 49s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/15/check/gate-nova-docs/228bd65/doc/build/html/ : SUCCESS in 4m 24s\n- gate-nova-python26 http://logs.openstack.org/75/56075/15/check/gate-nova-python26/4574080 : SUCCESS in 21m 37s\n- gate-nova-python27 http://logs.openstack.org/75/56075/15/check/gate-nova-python27/1c4d3e0 : SUCCESS in 15m 54s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/15/check/check-tempest-devstack-vm-full/9ebcc95 : SUCCESS in 31m 00s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/15/check/check-tempest-devstack-vm-postgres-full/3a631a5 : SUCCESS in 29m 42s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/15/check/check-tempest-devstack-vm-neutron/940f986 : SUCCESS in 24m 37s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/15/check/gate-tempest-devstack-vm-large-ops/8c9f087 : SUCCESS in 14m 11s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/15/check/gate-tempest-devstack-vm-neutron-large-ops/362e56a : SUCCESS in 17m 27s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/15/check/check-grenade-devstack-vm/3723072 : SUCCESS in 32m 17s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/15/check/gate-devstack-vm-cells/04667fa : SUCCESS in 8m 50s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/15/check/gate-nova-pylint/2fbbe10 : SUCCESS in 14m 11s (non-voting)\n","accounts_in_message":[],"_revision_number":15},{"id":"fc47591c2152d010dafcd5d0a6c03c0d06b9eeaf","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2013-11-14 03:57:29.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9a8eb1d605db468aa9e11d5877d903bca3eecc53","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-11-14 03:57:47.000000000","message":"Patch Set 16:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":16},{"id":"52a99cb12b0685ce6e9e443f25ead4ef02ff0a69","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-11-14 04:24:29.000000000","message":"Patch Set 16: Works for me\n\nSmokeStack Results (patch set 16):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/157066 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/157067 : SUCCESS \n","accounts_in_message":[],"_revision_number":16},{"id":"b5935556aac59402e9f372d259bb849bb718787e","author":{"_account_id":5292,"name":"Christopher Yeoh","email":"cbkyeoh@gmail.com","username":"cyeoh-0"},"date":"2013-11-14 04:27:05.000000000","message":"Patch Set 16: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":16},{"id":"e85f49223583dd998c5dd4336387703e5f2dc515","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 04:29:51.000000000","message":"Patch Set 16: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/75/56075/16/check/gate-nova-pep8/7386ba1 : SUCCESS in 3m 18s\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/16/check/gate-nova-docs/425681f/doc/build/html/ : SUCCESS in 3m 31s\n- gate-nova-python26 http://logs.openstack.org/75/56075/16/check/gate-nova-python26/13688ec : SUCCESS in 22m 58s\n- gate-nova-python27 http://logs.openstack.org/75/56075/16/check/gate-nova-python27/738884d : SUCCESS in 18m 03s\n- check-tempest-devstack-vm-full http://logs.openstack.org/75/56075/16/check/check-tempest-devstack-vm-full/9c8f61f : SUCCESS in 31m 41s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/16/check/check-tempest-devstack-vm-postgres-full/0e44af6 : SUCCESS in 32m 00s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/16/check/check-tempest-devstack-vm-neutron/e929738 : SUCCESS in 28m 40s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/16/check/gate-tempest-devstack-vm-large-ops/1a86b1b : SUCCESS in 13m 45s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/16/check/gate-tempest-devstack-vm-neutron-large-ops/1c3dfd7 : SUCCESS in 15m 52s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/75/56075/16/check/check-grenade-devstack-vm/da69ce4 : SUCCESS in 31m 34s\n- gate-devstack-vm-cells http://logs.openstack.org/75/56075/16/check/gate-devstack-vm-cells/0db7379 : SUCCESS in 10m 36s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/75/56075/16/check/gate-nova-pylint/3f2f3e4 : SUCCESS in 14m 41s (non-voting)\n","accounts_in_message":[],"_revision_number":16},{"id":"4dd2fe461b1672b9e39004deb5368b0271595d69","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2013-11-14 05:29:49.000000000","message":"Patch Set 16: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":16},{"id":"06b7da91153e8b49aacf8504ebb7d59faf7132a7","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2013-11-14 05:38:17.000000000","message":"Patch Set 16: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":16},{"id":"624b31ff5f5bcc8574909ffe098f74cfd4013eef","author":{"_account_id":1501,"name":"Phil Day","email":"philip.day@hp.com","username":"philip-day"},"date":"2013-11-14 14:10:00.000000000","message":"Patch Set 16: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":16},{"id":"2a938dfb46a3820a2a76ab7d426cc66c34e99480","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2013-11-14 20:27:46.000000000","message":"Patch Set 16: Looks good to me (core reviewer); Approved\n\nLGTM, this is one in many steps in making nova+neutron usable","accounts_in_message":[],"_revision_number":16},{"id":"938abc48ef45fa90858e660c207362e45c9b484d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 20:27:49.000000000","message":"Patch Set 16: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":16},{"id":"549f97d046d9bff0ef09d1be540cda2ce37663a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 23:35:18.000000000","message":"Patch Set 16: Verified\n\nBuild succeeded.\n\n- gate-nova-docs http://docs-draft.openstack.org/75/56075/16/gate/gate-nova-docs/501a803/doc/build/html/ : SUCCESS in 3m 32s\n- gate-nova-pep8 http://logs.openstack.org/75/56075/16/gate/gate-nova-pep8/d932ece : SUCCESS in 5m 42s\n- gate-nova-python26 http://logs.openstack.org/75/56075/16/gate/gate-nova-python26/31319b8 : SUCCESS in 23m 32s\n- gate-nova-python27 http://logs.openstack.org/75/56075/16/gate/gate-nova-python27/72c679d : SUCCESS in 18m 26s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/75/56075/16/gate/gate-tempest-devstack-vm-full/b1a21ce : SUCCESS in 31m 37s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/75/56075/16/gate/gate-tempest-devstack-vm-postgres-full/9b217de : SUCCESS in 33m 02s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/75/56075/16/gate/gate-tempest-devstack-vm-neutron/7cf895e : SUCCESS in 26m 33s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/75/56075/16/gate/gate-tempest-devstack-vm-large-ops/ab1c687 : SUCCESS in 19m 08s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/75/56075/16/gate/gate-tempest-devstack-vm-neutron-large-ops/623c18b : SUCCESS in 17m 51s (non-voting)\n- gate-grenade-devstack-vm http://logs.openstack.org/75/56075/16/gate/gate-grenade-devstack-vm/10d9ec5 : SUCCESS in 32m 48s\n","accounts_in_message":[],"_revision_number":16},{"id":"b0f0a3dec92eb15999fe7c669f326ae72352c359","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-14 23:35:20.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":16},{"id":"abfc8f43c6a097b2cefd985c3c892b21ae634897","author":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"date":"2013-11-15 23:10:03.000000000","message":"Patch Set 16: Reverted\n\nThis patchset was reverted in change: I6aec143311b4ae41c3dcf4147bd47bfb1afd15f9","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"85332012dede96fa6729026c2a90594ea0502ac5","revisions":{"015e960b2d97625f13a5f8e3bf1cad7a90685094":{"kind":"REWORK","_number":1,"created":"2013-11-12 18:28:33.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/1"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/015e960b2d97625f13a5f8e3bf1cad7a90685094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/015e960b2d97625f13a5f8e3bf1cad7a90685094"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bde3b555e7de87f149134df6928449f8c1b8a936":{"kind":"REWORK","_number":2,"created":"2013-11-12 19:19:50.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/2"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 19:19:27.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP for auth-plugins gets implemented\nhttps://blueprints.launchpad.net/python-keystoneclient/+spec/auth-plugins\nThat BP will ensure that tokens can be shared across clients as\nwell\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bde3b555e7de87f149134df6928449f8c1b8a936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bde3b555e7de87f149134df6928449f8c1b8a936"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4ba4c94ed3e6594cf8f7c0375b0bcf6ccb010d70":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2013-11-12 19:23:56.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/3"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 19:22:46.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP auth-plugins gets implemented.  That blue print will\nensure that tokens can be shared across clients as well\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ba4c94ed3e6594cf8f7c0375b0bcf6ccb010d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ba4c94ed3e6594cf8f7c0375b0bcf6ccb010d70"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2fca8a0c7b63caff03ea525134e30fabf4c1c285":{"kind":"REWORK","_number":4,"created":"2013-11-12 20:56:52.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/4"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 20:56:30.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP auth-plugins gets implemented.  That blue print will\nensure that tokens can be shared across clients as well\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fca8a0c7b63caff03ea525134e30fabf4c1c285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fca8a0c7b63caff03ea525134e30fabf4c1c285"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"004f5307b39afabc862bcba56f7687ce88761b12":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2013-11-12 20:56:52.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/5"}}},"commit":{"parents":[{"commit":"30fa37e7776831d6f8022f52d3d92f62189fb702","subject":"Moved quota headroom calculations into quota_reserve","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30fa37e7776831d6f8022f52d3d92f62189fb702"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-11-12 22:15:06.000000000","tz":0},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP auth-plugins gets implemented.  That blue print will\nensure that tokens can be shared across clients as well\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/004f5307b39afabc862bcba56f7687ce88761b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/004f5307b39afabc862bcba56f7687ce88761b12"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"30fa37e7776831d6f8022f52d3d92f62189fb702","is_merged_in_target_branch":true,"change_id":"I855b0cad495182c5c4ee42b2acf5eedd198557b5","change_number":51263,"patch_set_number":17,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bd7c3cbecb83b9c622e71643c1a35e75750e5055":{"kind":"REWORK","_number":6,"created":"2013-11-13 01:03:47.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/6"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 01:03:23.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP auth-plugins gets implemented.  That blue print will\nensure that tokens can be shared across clients as well\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd7c3cbecb83b9c622e71643c1a35e75750e5055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd7c3cbecb83b9c622e71643c1a35e75750e5055"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d9b42ed49859616d71c75cda02c80b878383a652":{"kind":"REWORK","_number":7,"created":"2013-11-13 01:33:58.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/7"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 01:33:35.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nNOTE: Some of the hackiness in this code will go away once\nwe have the BP auth-plugins gets implemented.  That blue print will\nensure that tokens can be shared across clients as well\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9b42ed49859616d71c75cda02c80b878383a652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9b42ed49859616d71c75cda02c80b878383a652"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6d93d43071119973dfb2a71f734fb89caf7bdc10":{"kind":"REWORK","_number":8,"created":"2013-11-13 03:30:57.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/8"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 03:30:19.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d93d43071119973dfb2a71f734fb89caf7bdc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d93d43071119973dfb2a71f734fb89caf7bdc10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"deca1a7f1f917b7e22dfb75c6bfa19a8571db776":{"kind":"REWORK","_number":9,"created":"2013-11-13 12:41:40.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/9"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 12:41:13.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/deca1a7f1f917b7e22dfb75c6bfa19a8571db776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/deca1a7f1f917b7e22dfb75c6bfa19a8571db776"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4374c686d58c0750f3fdfa1ea8513474c2199808":{"kind":"REWORK","_number":10,"created":"2013-11-13 13:25:54.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/10"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 13:25:24.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4374c686d58c0750f3fdfa1ea8513474c2199808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4374c686d58c0750f3fdfa1ea8513474c2199808"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"134d5b7b7970bc7ea5b821ff2239fc38814913b3":{"kind":"REWORK","_number":11,"created":"2013-11-13 19:53:04.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/11"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 19:52:38.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/134d5b7b7970bc7ea5b821ff2239fc38814913b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/134d5b7b7970bc7ea5b821ff2239fc38814913b3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6574b39536e0e74988741831288acff315e36333":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2013-11-13 19:53:04.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/12"}}},"commit":{"parents":[{"commit":"4bc19d5e14f112d50229902d5c6806e6f52f8515","subject":"Merge \"Make v3 admin_password parameters consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bc19d5e14f112d50229902d5c6806e6f52f8515"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-11-13 19:55:03.000000000","tz":0},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6574b39536e0e74988741831288acff315e36333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6574b39536e0e74988741831288acff315e36333"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4bc19d5e14f112d50229902d5c6806e6f52f8515","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8ae729d38fac52452c4dbadca3decd4cad13f4bd":{"kind":"REWORK","_number":13,"created":"2013-11-13 22:38:00.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/13"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-13 22:37:35.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ae729d38fac52452c4dbadca3decd4cad13f4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ae729d38fac52452c4dbadca3decd4cad13f4bd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"af983bdd4a42bee27b8bf51b96d665e0a87a284f":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2013-11-13 22:38:00.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/14"}}},"commit":{"parents":[{"commit":"0f3776d036be5c2174eb848af1e312f7df35de26","subject":"Merge \"Xenapi: Allow windows builds with xentools 6.1 and 6.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f3776d036be5c2174eb848af1e312f7df35de26"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-11-13 23:50:44.000000000","tz":0},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af983bdd4a42bee27b8bf51b96d665e0a87a284f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af983bdd4a42bee27b8bf51b96d665e0a87a284f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0f3776d036be5c2174eb848af1e312f7df35de26","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"25d258f0ad42a3b8797409a22275182c7039d63a":{"kind":"REWORK","_number":15,"created":"2013-11-14 01:10:05.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/15"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-14 01:09:38.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25d258f0ad42a3b8797409a22275182c7039d63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25d258f0ad42a3b8797409a22275182c7039d63a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"85332012dede96fa6729026c2a90594ea0502ac5":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2013-11-14 03:57:29.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/75/56075/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/56075/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/56075/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/56075/16"}}},"commit":{"parents":[{"commit":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","subject":"Merge \"xenapi: Move settings to their own config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65"}]}],"author":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-12 18:25:42.000000000","tz":-300},"committer":{"name":"Davanum Srinivas","email":"dims@linux.vnet.ibm.com","date":"2013-11-14 03:56:51.000000000","tz":-300},"subject":"Cache Neutron Client for Admin Scenarios","message":"Cache Neutron Client for Admin Scenarios\n\nStore a thread local copy of Neutron client for use with admin\ntokens, this will prevent excessive calls to keystone to mint\nnew tokens and will enable reuse of the already set up client\nfor subsequent calls in this thread.\n\nChange-Id: I6346281ca3516c31b195ce71942786d724e5e9df\nCloses-Bug: #1250580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85332012dede96fa6729026c2a90594ea0502ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85332012dede96fa6729026c2a90594ea0502ac5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d9692ec5ce2c2a01c9f8f49872072cd7cdfa2e65","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
