)]}'
{"id":"openstack%2Fnova~570018","triplet_id":"openstack%2Fnova~master~Iacff300b117a8ed629b92e53a6fe0651719eeaa9","project":"openstack/nova","branch":"master","topic":"bp/bandwidth-resource-provider","hashtags":[],"change_id":"Iacff300b117a8ed629b92e53a6fe0651719eeaa9","subject":"Remove network info stubbing in functional test","status":"MERGED","created":"2018-05-22 18:06:13.000000000","updated":"2018-06-22 03:08:47.000000000","submitted":"2018-06-22 03:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"570018-1529636927239-5a47a813","meta_rev_id":"e014f3d2a1fecd98a80ec7539635173ebe06b68b","_number":570018,"virtual_id_number":570018,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-06-13 17:31:19.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-19 23:33:55.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"date":"2018-06-13 13:53:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-06-22 03:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-13 12:06:45.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"date":"2018-06-13 13:03:21.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-13 15:18:48.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"date":"2018-06-14 03:58:55.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-06-21 17:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2018-06-18 15:56:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-14 13:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":1,"date":"2018-06-21 14:43:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2018-06-15 22:36:02.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-21 17:23:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-28 12:15:26.000000000","updated_by":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"reviewer":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"state":"REVIEWER"},{"updated":"2018-06-12 02:04:02.000000000","updated_by":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"reviewer":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"state":"REVIEWER"},{"updated":"2018-06-12 07:54:01.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2018-06-13 09:17:05.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2018-06-13 13:03:21.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-13 13:53:31.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-06-13 15:18:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-13 17:31:19.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-14 03:58:55.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-06-14 13:55:42.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-06-15 22:36:02.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2018-06-18 15:56:26.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-06-19 23:33:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-21 14:43:42.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2018-06-21 17:23:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-06-22 03:08:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9013346c39d3aec89df945fd7d77b5f3486ef9b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-22 18:06:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"816d8d1893c8319debdd66869b31985037f7c788","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-22 18:51:48.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/1/check-vote/ext-nova-zuul/86e579a : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":1},{"id":"11e6a8554bc732175e8666f28700b4926fa66843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-22 19:39:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/1/check/openstack-tox-pep8/9c63428/ : SUCCESS in 9m 41s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/1/check/openstack-tox-py27/64afbd4/ : SUCCESS in 10m 44s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/1/check/openstack-tox-py35/2c0c498/ : SUCCESS in 10m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/1/check/build-openstack-sphinx-docs/44c16bb/html/ : SUCCESS in 6m 06s\n- tempest-full http://logs.openstack.org/18/570018/1/check/tempest-full/29d17fc/ : SUCCESS in 1h 25m 39s\n- neutron-grenade http://logs.openstack.org/18/570018/1/check/neutron-grenade/0427fc1/ : SUCCESS in 58m 32s\n- tempest-full-py3 http://logs.openstack.org/18/570018/1/check/tempest-full-py3/0959b1f/ : SUCCESS in 1h 27m 03s\n- nova-tox-functional http://logs.openstack.org/18/570018/1/check/nova-tox-functional/eca9929/ : SUCCESS in 15m 03s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/1/check/nova-tox-functional-py35/decd920/ : SUCCESS in 15m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/1/check/openstack-tox-lower-constraints/983a5db/ : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0dd861b4123f397bf3483d74c58a41f42ee2498a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-05-22 19:41:33.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-18954 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ae19c50f4432c89c7f593c6666aead4442c953f6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-05-22 20:14:15.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-570018-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"8f5556c7615d31329fc7cf611443d988810e6689","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-05-22 22:57:21.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/570018/1/ : FAILURE in 4h 50m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ce2ef7d06aae6e63213fb7aab85996631644fbea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-05-24 22:06:03.000000000","message":"Patch Set 1: Code-Review+1\n\nSoft approve pending a look at the rest of the series.","accounts_in_message":[],"_revision_number":1},{"id":"efef6feb9b5169ce789514ef977065cfa546d092","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-05-28 02:30:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"76be39c51047a3e76f7ad27a9b5468c86c140aae","author":{"_account_id":26311,"name":"tianhui","email":"tianhui@awcloud.com","username":"tianhui"},"date":"2018-05-28 12:15:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8a227d39fab8f8e65cbdda88f1ba912ed0128cdc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-05-29 07:07:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b6f3b282abac87c6fd7c8b61626075665f68fedf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-05-29 07:08:19.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19089 : FAILURE in 26s","accounts_in_message":[],"_revision_number":2},{"id":"0e3e1d64fa80b23c4f0692f6243e2174bccc33a9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-29 07:49:28.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/2/check-vote/ext-nova-zuul/e01e94c : FAILURE in 41m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f235aacd47e2484c9d3e4c37fde4a49a52b43b13","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-05-29 08:25:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/570018/2/check/nova-out-of-tree-pvm/a49445e : SUCCESS in 1h 17m 30s\n- nova-in-tree-pvm http://184.172.12.213/18/570018/2/check/nova-in-tree-pvm/14a13c7 : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2b5a786cc76878c1827832422c000525f5738f3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-29 08:58:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/2/check/openstack-tox-pep8/200747a/ : SUCCESS in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/2/check/openstack-tox-py27/b0a2485/ : SUCCESS in 12m 56s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/2/check/openstack-tox-py35/868a1af/ : SUCCESS in 11m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/2/check/build-openstack-sphinx-docs/41f2da3/html/ : SUCCESS in 6m 23s\n- tempest-full http://logs.openstack.org/18/570018/2/check/tempest-full/7d10921/ : SUCCESS in 1h 26m 56s\n- neutron-grenade http://logs.openstack.org/18/570018/2/check/neutron-grenade/afbffed/ : FAILURE in 4m 22s\n- tempest-full-py3 http://logs.openstack.org/18/570018/2/check/tempest-full-py3/295c2ae/ : SUCCESS in 1h 49m 24s\n- nova-tox-functional http://logs.openstack.org/18/570018/2/check/nova-tox-functional/190ac07/ : SUCCESS in 13m 03s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/2/check/nova-tox-functional-py35/fc8d359/ : SUCCESS in 22m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/2/check/openstack-tox-lower-constraints/81033f0/ : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0c7526e0e118c71fda340789cc98cab63be77f33","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-05-29 11:49:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/570018/2/ : SUCCESS in 4h 41m 38s","accounts_in_message":[],"_revision_number":2},{"id":"caa29c35f3b261988f83a32c0cd7df49f2c0fb07","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-01 20:59:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a70b37654a9ab7300501e8e438fb57c387774487","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-06 11:32:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7cffa41b7f99ea47a1276562cffc019f8b23e2be","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-06 12:25:40.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/3/check-vote/ext-nova-zuul/a21290a : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":3},{"id":"864e91df43445ed4760329e24ba550237b8454b7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-06 12:58:10.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/570018/3/check/nova-out-of-tree-pvm/2291a28 : SUCCESS in 1h 20m 06s\n- nova-in-tree-pvm http://184.172.12.213/18/570018/3/check/nova-in-tree-pvm/b4539d4 : FAILURE in 1h 24m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2cea59d90e19818cc835cd96ceab6de2a1d05759","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-06 13:08:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/3/check/openstack-tox-pep8/637bc36/ : SUCCESS in 11m 29s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/3/check/openstack-tox-py27/0c2dcb8/ : SUCCESS in 11m 30s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/3/check/openstack-tox-py35/2c13b96/ : SUCCESS in 12m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/3/check/build-openstack-sphinx-docs/2de18b0/html/ : SUCCESS in 6m 47s\n- tempest-full http://logs.openstack.org/18/570018/3/check/tempest-full/1090d8f/ : SUCCESS in 1h 23m 24s\n- neutron-grenade http://logs.openstack.org/18/570018/3/check/neutron-grenade/3787479/ : SUCCESS in 46m 57s\n- tempest-full-py3 http://logs.openstack.org/18/570018/3/check/tempest-full-py3/52dfe96/ : SUCCESS in 1h 29m 38s\n- nova-tox-functional http://logs.openstack.org/18/570018/3/check/nova-tox-functional/a61445c/ : SUCCESS in 14m 53s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/3/check/nova-tox-functional-py35/f2e86d1/ : SUCCESS in 15m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/3/check/openstack-tox-lower-constraints/4b295a1/ : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":3},{"id":"96b38708994e229e4a06c85f02ddd08b46976aae","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-06 13:24:14.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19443 : FAILURE in 1h 50m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f803132b74d9338c11ea2028258f278e1908262b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-06 14:00:01.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-570018-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"0325897c341c1b18b9e284e1a8bf63619f5cc085","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-06 14:01:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/570018/3/ : SUCCESS in 2h 28m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d522fbf0a6d55bd032803a7ff69a5e6303c69371","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2018-06-07 04:46:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"75582c12554e5ac9a275e46ad3263426f1da9fa3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-06-09 00:55:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/570018/3/EMC_ScaleIO/1002 : SUCCESS in 1h 30m 58s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"34f3411a393100ba9b9cde1b35f8154d2c30edb0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-11 07:31:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"731b8ace3b21a337d0259996a5ceb0bec08dacb5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-11 08:28:36.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/4/check-vote/ext-nova-zuul/edd0ffc : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":4},{"id":"05f9cfee2dbb1fe4c6879e77b54f2d040224fe81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-11 09:07:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/4/check/openstack-tox-pep8/bfbcd2a/ : SUCCESS in 10m 47s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/4/check/openstack-tox-py27/363c855/ : SUCCESS in 15m 36s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/4/check/openstack-tox-py35/e6a2998/ : SUCCESS in 12m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/4/check/build-openstack-sphinx-docs/b13cb09/html/ : SUCCESS in 7m 01s\n- tempest-full http://logs.openstack.org/18/570018/4/check/tempest-full/bac3390/ : SUCCESS in 1h 33m 21s\n- neutron-grenade http://logs.openstack.org/18/570018/4/check/neutron-grenade/37fcd84/ : SUCCESS in 54m 25s\n- tempest-full-py3 http://logs.openstack.org/18/570018/4/check/tempest-full-py3/92e420f/ : SUCCESS in 1h 20m 41s\n- nova-tox-functional http://logs.openstack.org/18/570018/4/check/nova-tox-functional/655f7ca/ : FAILURE in 16m 23s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/4/check/nova-tox-functional-py35/94cb864/ : FAILURE in 15m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/4/check/openstack-tox-lower-constraints/7bfe9a4/ : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":4},{"id":"d47c6be85a8bb85b7f31b72fa6c6eacdfe445bf6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-11 09:17:47.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19696 : SUCCESS in 1h 45m 56s","accounts_in_message":[],"_revision_number":4},{"id":"0c3b821b291dc9b6572d96e73461b07111c717d2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-11 11:50:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/570018/4/ : SUCCESS in 3h 11m 24s","accounts_in_message":[],"_revision_number":4},{"id":"6d95e5a7ec96f76be2b295fc6a9a238128f070db","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-11 14:27:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a280e9aacfba8e9c0f4078f91e1069656a122d07","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-11 14:29:18.000000000","message":"Patch Set 5:\n\nI needed to fix up a new set of tests after the rebase.\ngibi","accounts_in_message":[],"_revision_number":5},{"id":"6ab7f760544758cf8395124aad55aba5a9989710","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-11 15:23:42.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/5/check-vote/ext-nova-zuul/5bebb85 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":5},{"id":"4e90fa58d6e4c8ea68fac3e407a8ec5d40cda2d2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-11 15:49:25.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19717 : FAILURE in 1h 21m 02s","accounts_in_message":[],"_revision_number":5},{"id":"46289146e311ad2c665a0d00735c27a201b54df1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-11 15:55:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/5/check/openstack-tox-pep8/183ecc7/ : SUCCESS in 10m 57s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/5/check/openstack-tox-py27/7ed06cd/ : SUCCESS in 13m 55s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/5/check/openstack-tox-py35/618f43b/ : SUCCESS in 18m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/5/check/build-openstack-sphinx-docs/00bd6e1/html/ : SUCCESS in 7m 02s\n- tempest-full http://logs.openstack.org/18/570018/5/check/tempest-full/b07a20c/ : SUCCESS in 1h 24m 32s\n- neutron-grenade http://logs.openstack.org/18/570018/5/check/neutron-grenade/235fca6/ : SUCCESS in 52m 23s\n- tempest-full-py3 http://logs.openstack.org/18/570018/5/check/tempest-full-py3/d8c550e/ : SUCCESS in 1h 21m 02s\n- nova-tox-functional http://logs.openstack.org/18/570018/5/check/nova-tox-functional/68f36a6/ : SUCCESS in 16m 12s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/5/check/nova-tox-functional-py35/2151eb5/ : SUCCESS in 18m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/5/check/openstack-tox-lower-constraints/a23a297/ : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":5},{"id":"ff2ec5e8be26b0cada06e74fce900151cfc2c7f5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-11 16:48:46.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-570018-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"c75ddd81211cd31835d2e410e1d8e92145318a10","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-11 18:33:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bcab4e3e84ce506bed84ee4297b36e4005da3a8e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-11 20:05:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/570018/5/ : SUCCESS in 3h 18m 09s","accounts_in_message":[],"_revision_number":5},{"id":"10bde6764b54d47173d9633a56276a4226d7a42f","author":{"_account_id":27386,"name":"gaofei","email":"gao.fei@inspur.com","username":"gaofei"},"date":"2018-06-12 02:04:02.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cce511a6e50699850b726afcc32e00a27c61bcc7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-12 02:36:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/570018/5/check/nova-out-of-tree-pvm/11f5684 : SUCCESS in 2h 32m 52s\n- nova-in-tree-pvm http://184.172.12.213/18/570018/5/check/nova-in-tree-pvm/498cfc5 : SUCCESS in 2h 13m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b567503f5e3d589677426b06c7e69aec69932260","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2018-06-12 07:54:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c256de058f84da7956a4945c7abfeaf851e0132d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-12 13:55:03.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c7bd066987bb1ec37954d5625d1014472a51e717","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-12 15:02:22.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/6/check-vote/ext-nova-zuul/3e49a36 : FAILURE in 1h 06m 32s","accounts_in_message":[],"_revision_number":6},{"id":"e28e1fff7f8dd66272af07df3cde6969fb67ed10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-12 15:34:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/6/check/openstack-tox-pep8/3b3d2ef/ : SUCCESS in 10m 19s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/6/check/openstack-tox-py27/ba6a41c/ : SUCCESS in 12m 43s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/6/check/openstack-tox-py35/31470b2/ : SUCCESS in 17m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/6/check/build-openstack-sphinx-docs/54b619c/html/ : SUCCESS in 7m 23s\n- tempest-full-py3 http://logs.openstack.org/18/570018/6/check/tempest-full-py3/414bb01/ : SUCCESS in 1h 14m 25s\n- nova-tox-functional http://logs.openstack.org/18/570018/6/check/nova-tox-functional/5b758a0/ : SUCCESS in 23m 17s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/6/check/nova-tox-functional-py35/bad8162/ : SUCCESS in 16m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/6/check/openstack-tox-lower-constraints/550d2a0/ : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":6},{"id":"2e3a3b5bfc491ffae2a869fe878d092e8494507c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-12 18:05:33.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/570018/6/check/nova-out-of-tree-pvm/385d647 : SUCCESS in 2h 18m 36s\n- nova-in-tree-pvm http://184.172.12.213/18/570018/6/check/nova-in-tree-pvm/d773f82 : FAILURE in 2h 39m 19s","accounts_in_message":[],"_revision_number":6},{"id":"ef9204ce07bdc1d9a28ea915ac2c412d776534e1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-12 18:08:53.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-570018-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"39aba1ed35f0dffbaa71044af7259869ac485a78","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-13 04:32:42.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/570018/6/ : FAILURE in 4h 38m 24s","accounts_in_message":[],"_revision_number":6},{"id":"bbf61cfac33c07b392c8ca9d1880957372b9d5ca","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2018-06-13 09:17:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4671edc4d59db8f593cd69790ee4674ce03cb18e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-13 12:05:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e83aa3244aafcab8c39417aab62079bb514543e5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-13 12:06:45.000000000","message":"Patch Set 7:\n\nI had to resolve merge conflict in the whole series.","accounts_in_message":[],"_revision_number":7},{"id":"a776330fd451a7045d3275ddb8f57bd7ad34c968","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-13 13:03:21.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/18/570018/7/check-vote/ext-nova-zuul/cb7d7d1 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":7},{"id":"77fb549ae7d17d78244d1e6b8ddb54307d02966d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 13:43:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/7/check/openstack-tox-pep8/013ac3b/ : SUCCESS in 10m 10s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/7/check/openstack-tox-py27/2076f86/ : SUCCESS in 14m 01s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/7/check/openstack-tox-py35/3b5d984/ : SUCCESS in 12m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/7/check/build-openstack-sphinx-docs/f149d90/html/ : SUCCESS in 7m 27s\n- tempest-full-py3 http://logs.openstack.org/18/570018/7/check/tempest-full-py3/1cc715f/ : SUCCESS in 1h 23m 09s\n- nova-tox-functional http://logs.openstack.org/18/570018/7/check/nova-tox-functional/7f09df5/ : SUCCESS in 21m 31s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/7/check/nova-tox-functional-py35/cba8680/ : SUCCESS in 16m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/7/check/openstack-tox-lower-constraints/6bf25fe/ : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":7},{"id":"ea41cb6df7708854f99e0e1d7ad2b1699179f470","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-13 13:53:31.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19901 : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":7},{"id":"c35869f02214a218159daea92daf087913baa967","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-13 15:18:48.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-18-570018-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"06288d571ca25766e6247f134570299e7f88ea5f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-13 17:31:19.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/18/570018/7/check/nova-out-of-tree-pvm/adbfea0 : FAILURE in 5h 14m 45s\n- nova-in-tree-pvm http://184.172.12.213/18/570018/7/check/nova-in-tree-pvm/15be673 : FAILURE in 5h 16m 31s","accounts_in_message":[],"_revision_number":7},{"id":"5e5476021828621d2ceab26b8a01d84569ab3562","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-14 03:58:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/570018/7/ : SUCCESS in 2h 36m 00s","accounts_in_message":[],"_revision_number":7},{"id":"e6818fe84af8d432c4ce612cdc3ba7819ce9fbb6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-06-14 13:55:42.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"20a3d4c6108027ac2f945c454775afab43e8d418","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-06-15 22:36:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2aaa8470f10db6dba47ab4a115ad762471abf682","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-06-18 15:56:26.000000000","message":"Patch Set 7: Code-Review+1\n\nyeah, good sense, test only what we\u0027re testing","accounts_in_message":[],"_revision_number":7},{"id":"76217cc97ebdd965107023bbe8f94bf489902149","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-06-19 23:33:55.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_ScaleIO http://publiclogs.emc.com/570018/7/EMC_ScaleIO/1165 : FAILURE in 1h 59m 33s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":7},{"id":"a7fe3e4ed1f3968df6cd2ed42c5c4b998ee07ed8","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-06-21 14:43:42.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"684ba53df167cf218a5f1b58dd35b21e4628b352","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-21 17:23:27.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nYup, this makes sense even in an independent context","accounts_in_message":[],"_revision_number":7},{"id":"547e83d4db68f8b74ba37e7b15302357cfb0e114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 17:24:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"60cb6292a9018acef37315bdfeb3af95ee38d982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 03:08:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/570018/7/gate/openstack-tox-pep8/3bdc9f0/ : SUCCESS in 10m 19s\n- openstack-tox-py27 http://logs.openstack.org/18/570018/7/gate/openstack-tox-py27/6a1443c/ : SUCCESS in 10m 34s\n- openstack-tox-py35 http://logs.openstack.org/18/570018/7/gate/openstack-tox-py35/079e04d/ : SUCCESS in 12m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/18/570018/7/gate/build-openstack-sphinx-docs/d370279/html/ : SUCCESS in 7m 25s\n- tempest-full-py3 http://logs.openstack.org/18/570018/7/gate/tempest-full-py3/9b65a0e/ : SUCCESS in 1h 28m 44s\n- nova-tox-functional http://logs.openstack.org/18/570018/7/gate/nova-tox-functional/2fea6ab/ : SUCCESS in 14m 59s\n- nova-tox-functional-py35 http://logs.openstack.org/18/570018/7/gate/nova-tox-functional-py35/f663e53/ : SUCCESS in 13m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/570018/7/gate/openstack-tox-lower-constraints/dc8d969/ : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":7},{"id":"a82bbe56efbb9c452fc312d9c0efbf9e18a5a53a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 03:08:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8430e31fab9aaab687591e411882358997da1b3d","revisions":{"aad1dc54f883338b7c7ca3f48c7301de4fc863fb":{"kind":"REWORK","_number":1,"created":"2018-05-22 18:06:13.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/1"}}},"commit":{"parents":[{"commit":"20b99f6998c088650b0c0cb066cc6aac3e5f9312","subject":"Merge \"Granular requests to get_allocation_candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20b99f6998c088650b0c0cb066cc6aac3e5f9312"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 16:48:18.000000000","tz":-420},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad1dc54f883338b7c7ca3f48c7301de4fc863fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad1dc54f883338b7c7ca3f48c7301de4fc863fb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"20b99f6998c088650b0c0cb066cc6aac3e5f9312","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a11fefc4406cab698b31fd89e4a9e1cbc1372f76":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-05-29 07:07:21.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/2"}}},"commit":{"parents":[{"commit":"5bf9aa7fd455e93b73dc13aba66ccb7ec00f23a5","subject":"Merge \"Change exception type while deattaching root device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bf9aa7fd455e93b73dc13aba66ccb7ec00f23a5"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-29 06:55:13.000000000","tz":-420},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a11fefc4406cab698b31fd89e4a9e1cbc1372f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a11fefc4406cab698b31fd89e4a9e1cbc1372f76"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5bf9aa7fd455e93b73dc13aba66ccb7ec00f23a5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c58ce88817373760b4f0ad1b17582b8140a33d13":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-06-06 11:32:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/3"}}},"commit":{"parents":[{"commit":"e6f8349a72627dcbb943f4c73b0d2e82834031f5","subject":"Merge \"Fix some inconsistencies in doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e6f8349a72627dcbb943f4c73b0d2e82834031f5"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-06-06 10:49:46.000000000","tz":120},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c58ce88817373760b4f0ad1b17582b8140a33d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c58ce88817373760b4f0ad1b17582b8140a33d13"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e6f8349a72627dcbb943f4c73b0d2e82834031f5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1231bcddc8644e2c32e7dbd9bd5ebd9fdd76a382":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-06-11 07:31:15.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/4"}}},"commit":{"parents":[{"commit":"ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649","subject":"Merge \"Remove unused _disk_qcow2_to_raw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-06-11 07:31:05.000000000","tz":120},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1231bcddc8644e2c32e7dbd9bd5ebd9fdd76a382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1231bcddc8644e2c32e7dbd9bd5ebd9fdd76a382"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4e2c32473d0de98182239a415f6e96820977aa96":{"kind":"REWORK","_number":5,"created":"2018-06-11 14:27:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/5"}}},"commit":{"parents":[{"commit":"ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649","subject":"Merge \"Remove unused _disk_qcow2_to_raw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-06-11 14:27:42.000000000","tz":120},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e2c32473d0de98182239a415f6e96820977aa96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e2c32473d0de98182239a415f6e96820977aa96"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ce3cfa0f89fae90c2de64cd8b4d5d48de29bb649","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"cf4b2ac7526b379bffa4776d124b9d64b63315f1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-06-12 13:55:03.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/6"}}},"commit":{"parents":[{"commit":"5ef23af397820ac2031731e0a370fdb1236d9563","subject":"Merge \"Mention running rootwrap in daemon mode if hitting vif plug timeouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ef23af397820ac2031731e0a370fdb1236d9563"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-06-12 10:56:25.000000000","tz":120},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cf4b2ac7526b379bffa4776d124b9d64b63315f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cf4b2ac7526b379bffa4776d124b9d64b63315f1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ef23af397820ac2031731e0a370fdb1236d9563","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8430e31fab9aaab687591e411882358997da1b3d":{"kind":"REWORK","_number":7,"created":"2018-06-13 12:05:32.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/18/570018/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/570018/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/570018/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/570018/7"}}},"commit":{"parents":[{"commit":"ca7d23a3e7287716781a042fe6038b876c0adaf4","subject":"Merge \"Do not use nova.test in placement.test_deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca7d23a3e7287716781a042fe6038b876c0adaf4"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-05-22 00:33:27.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@ericsson.com","date":"2018-06-13 11:58:38.000000000","tz":120},"subject":"Remove network info stubbing in functional test","message":"Remove network info stubbing in functional test\n\nThe ProviderUsageBaseTestCase stubs out lot of the port binding\nbehavior of nova to make the test pass with empty networks list.\nSince API microversion 2.37 nova allows passing \u0027none\u0027 as the networks\nparameter of server create to signal that no networking is needed for\nthe instance.\n\nThis patch proposes to use networks\u003d\u0027none\u0027 instead of stubbing the port\nbinding code path. This way later patches can introduce tests where\nthe port binding code path is also verified.\n\nIn separate future patches the usage of\nnova.tests.unit.fake_network.set_stub_network_methods can be eliminated\nfrom the functional tests.\n\nChange-Id: Iacff300b117a8ed629b92e53a6fe0651719eeaa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8430e31fab9aaab687591e411882358997da1b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8430e31fab9aaab687591e411882358997da1b3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ca7d23a3e7287716781a042fe6038b876c0adaf4","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
