)]}'
{"id":"openstack%2Fnova~571330","triplet_id":"openstack%2Fnova~master~I0cd5fc54d981acbf6a63ae74e272f83f24246c70","project":"openstack/nova","branch":"master","topic":"bp/mox-removal-stein","hashtags":[],"change_id":"I0cd5fc54d981acbf6a63ae74e272f83f24246c70","subject":"Remove mox in libvirt/test_driver.py (6)","status":"MERGED","created":"2018-05-30 23:20:09.000000000","updated":"2018-09-21 13:59:08.000000000","submitted":"2018-09-21 13:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"571330-1537538348955-ffe5684f","meta_rev_id":"adbcdf6d2460a82866a2ee2ed3d06c0534f6f081","_number":571330,"virtual_id_number":571330,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-09-18 05:54:02.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"date":"2018-09-18 02:33:43.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-09-18 02:36:40.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2018-09-18 08:28:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2018-09-21 13:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"date":"2018-09-18 21:45:58.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-09-18 02:40:20.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2018-09-18 04:19:30.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-18 02:09:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":2,"date":"2018-09-21 09:55:44.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-18 02:09:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2018-09-18 02:09:40.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-18 03:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":1,"date":"2018-09-21 09:55:44.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-25 09:58:59.000000000","updated_by":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"reviewer":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"state":"REVIEWER"},{"updated":"2018-08-01 14:43:46.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2018-08-05 00:16:57.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2018-08-06 03:17:11.000000000","updated_by":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"reviewer":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"state":"REVIEWER"},{"updated":"2018-08-23 17:19:44.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2018-08-26 23:42:39.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-29 03:36:28.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2018-09-03 11:38:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-09-18 02:33:43.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-18 02:36:40.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-18 02:40:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-18 03:44:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2018-09-18 04:19:30.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-09-18 05:54:02.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-18 08:28:06.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-09-18 21:45:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-21 09:55:44.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2018-09-21 13:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c162367631a7695edfc13f8c25d233d3c401bc65","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-05-30 23:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66033c7a101f2bb3739f1d192bcb32b36bffbcc9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-05-31 00:16:35.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/1/check-vote/ext-nova-zuul/8fcb240 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4571de0d2e0ccc20721f93444c0c9eae2aa33c01","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-05-31 00:27:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/1/check/nova-out-of-tree-pvm/a55dadb : SUCCESS in 1h 02m 49s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/1/check/nova-in-tree-pvm/4d62149 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"90d58f45def442176e3d0bb68aa94234dc920e9f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-05-31 00:39:01.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19182 : SUCCESS in 1h 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c07241f65be2c01a3271df80070b9f37c9228981","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-31 01:09:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/1/check/openstack-tox-pep8/bcfa65d/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/1/check/openstack-tox-py27/13b2bc6/ : SUCCESS in 11m 14s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/1/check/openstack-tox-py35/6d49234/ : SUCCESS in 11m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/1/check/build-openstack-sphinx-docs/b8646a7/html/ : SUCCESS in 6m 57s\n- tempest-full http://logs.openstack.org/30/571330/1/check/tempest-full/dad0e72/ : SUCCESS in 1h 41m 37s\n- neutron-grenade http://logs.openstack.org/30/571330/1/check/neutron-grenade/6518db1/ : SUCCESS in 47m 52s\n- tempest-full-py3 http://logs.openstack.org/30/571330/1/check/tempest-full-py3/91adbe7/ : SUCCESS in 1h 20m 12s\n- nova-tox-functional http://logs.openstack.org/30/571330/1/check/nova-tox-functional/9c4bd7a/ : SUCCESS in 15m 43s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/1/check/nova-tox-functional-py35/7f701f6/ : SUCCESS in 13m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/1/check/openstack-tox-lower-constraints/f5273ff/ : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ed516f21eb3804891fc3230d41f81466981c1ec6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-05-31 01:43:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/1/ : SUCCESS in 2h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c1a25b0bdd8ed0327cbe08cac52d5c63fd683444","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-05-31 01:55:46.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"ac5fcc2205e50112b2e72e354320f200bde245de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-06-02 00:58:40.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/571330/1/EMC_ScaleIO/894 : SUCCESS in 1h 26m 11s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":1},{"id":"7a40e2ada7ee1607cf2ebf86ce077bb5676d10fe","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-06-03 22:45:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1d2e3b9747762b5ff3c100d33b40b5484eefe612","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-03 23:37:43.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/2/check-vote/ext-nova-zuul/99c51cb : FAILURE in 51m 11s","accounts_in_message":[],"_revision_number":2},{"id":"904f9fb2424a72205000eafdd1c00390e6bcb1ae","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-04 00:00:44.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19304 : FAILURE in 1h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"0e7a2cb4bc830f1b5e7ff59a839e9579dd32b48d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-04 00:04:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/2/check/nova-out-of-tree-pvm/77cd825 : SUCCESS in 1h 16m 35s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/2/check/nova-in-tree-pvm/f8ce194 : SUCCESS in 1h 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7f489e12d0c387960cf1f7295bae0c696d8874c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-04 00:18:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/2/check/openstack-tox-pep8/141df1d/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/2/check/openstack-tox-py27/388f154/ : SUCCESS in 12m 09s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/2/check/openstack-tox-py35/4811d6e/ : SUCCESS in 10m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/2/check/build-openstack-sphinx-docs/b0e6e72/html/ : SUCCESS in 5m 50s\n- tempest-full http://logs.openstack.org/30/571330/2/check/tempest-full/60f5465/ : SUCCESS in 1h 30m 08s\n- neutron-grenade http://logs.openstack.org/30/571330/2/check/neutron-grenade/1b37e1c/ : SUCCESS in 49m 48s\n- tempest-full-py3 http://logs.openstack.org/30/571330/2/check/tempest-full-py3/265a529/ : SUCCESS in 1h 14m 12s\n- nova-tox-functional http://logs.openstack.org/30/571330/2/check/nova-tox-functional/2d9e10f/ : SUCCESS in 15m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/2/check/nova-tox-functional-py35/04fa483/ : SUCCESS in 13m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/2/check/openstack-tox-lower-constraints/5e34349/ : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":2},{"id":"db3c91a395d7879f329183d9d099db9012015649","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-04 00:52:21.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"c0d8f797518bf3d84d9308406222d6306d747ba9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-04 01:03:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/571330/2/ : FAILURE in 2h 17m 13s","accounts_in_message":[],"_revision_number":2},{"id":"cd11350df4117eb4b8377ac77cfd9c3cc6ceee27","author":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"date":"2018-06-04 02:00:40.000000000","message":"Patch Set 2:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":2},{"id":"dcd72b95bf4d18363204f8a226617bd675ff7131","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-04 05:47:18.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19309 : FAILURE in 3h 44m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1110333d99fbebc6db006b0687a3252ee27f8bbf","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-06-07 21:37:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"04240c3b33befb0d0ad45f4b69aff141d6ff8b90","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-07 22:28:29.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/3/check-vote/ext-nova-zuul/f75f47a : FAILURE in 47m 13s","accounts_in_message":[],"_revision_number":3},{"id":"669f9919f03aa84759f98f132076292f791932b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-07 23:20:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/3/check/openstack-tox-pep8/4156000/ : SUCCESS in 9m 41s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/3/check/openstack-tox-py27/05b52b5/ : SUCCESS in 11m 26s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/3/check/openstack-tox-py35/e81458a/ : SUCCESS in 11m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/3/check/build-openstack-sphinx-docs/e97de3e/html/ : SUCCESS in 7m 06s\n- tempest-full http://logs.openstack.org/30/571330/3/check/tempest-full/a612809/ : SUCCESS in 1h 34m 36s\n- neutron-grenade http://logs.openstack.org/30/571330/3/check/neutron-grenade/fb23fc9/ : SUCCESS in 53m 38s\n- tempest-full-py3 http://logs.openstack.org/30/571330/3/check/tempest-full-py3/aab9437/ : SUCCESS in 1h 30m 13s\n- nova-tox-functional http://logs.openstack.org/30/571330/3/check/nova-tox-functional/67e220b/ : SUCCESS in 14m 06s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/3/check/nova-tox-functional-py35/5647602/ : SUCCESS in 15m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/3/check/openstack-tox-lower-constraints/938b6b1/ : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8ee6efb475ccbd3ee28ba5a0a5d8cb58d6b3ebe6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-07 23:48:52.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-19579 : FAILURE in 2h 05m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2c7435c70e25c69e39f1ddb4b7880fbea38a0b75","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-08 02:19:35.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"b00b738ecba6b6968cda77a5f68a192376fc4cc6","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-08 04:45:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/3/check/nova-out-of-tree-pvm/5840248 : SUCCESS in 1h 09m 49s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/3/check/nova-in-tree-pvm/24db396 : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5d109cc28a1ef5a9fac63c015704941885d30b85","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-08 08:59:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/3/ : SUCCESS in 2h 46m 04s","accounts_in_message":[],"_revision_number":3},{"id":"aa96358eba19a0c48877e6cd4d3ec05f906a887a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-06-13 10:46:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/571330/3/EMC_ScaleIO/1066 : SUCCESS in 1h 37m 55s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":3},{"id":"d5418c080a164c0716dd82181c0b1299bc519d1b","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-06-22 04:31:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1d6ed080dda789bf6171c6121065d5d2606dd398","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-22 05:30:04.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/4/check-vote/ext-nova-zuul/89cc04b : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":4},{"id":"fb8e531cae3af18b37af94adfa1a6e6e528d8dc8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-22 06:04:45.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20537 : FAILURE in 1h 15m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ec5a6f25cc259583d0d455f82a2a003bb4b7f4a9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-06-22 06:25:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-06-22/571330/4/check/tempest-dsvm-intel-nfv-xenial/894672d : SUCCESS in 30m 25s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/571330/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/097506c : FAILURE in 1h 31m 09s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/571330/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/a17b32d : SUCCESS in 57m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a3b7765e18fb7c38cd5be8a37ff0b1951e74be0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 06:42:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/4/check/openstack-tox-pep8/5665641/ : SUCCESS in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/4/check/openstack-tox-py27/a55ceed/ : SUCCESS in 11m 08s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/4/check/openstack-tox-py35/b0d15ba/ : SUCCESS in 12m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/4/check/build-openstack-sphinx-docs/358fbc9/html/ : SUCCESS in 6m 10s\n- tempest-full-py3 http://logs.openstack.org/30/571330/4/check/tempest-full-py3/e967385/ : SUCCESS in 2h 07m 03s\n- nova-tox-functional http://logs.openstack.org/30/571330/4/check/nova-tox-functional/3b512ea/ : SUCCESS in 17m 23s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/4/check/nova-tox-functional-py35/74aa759/ : SUCCESS in 13m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/4/check/openstack-tox-lower-constraints/8cfc63d/ : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":4},{"id":"75053fc7820f5f1c3da48eec51e62413f5a5bfe6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-06-22 06:42:28.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"fee92e5e302f37e50e1b9b3a517d8c48194d42e1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-22 19:42:04.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/571330/4/ : FAILURE in 2h 49m 19s","accounts_in_message":[],"_revision_number":4},{"id":"c07c09d5aeb19e866d1dc19b8e4ee4f0f45dca62","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-06-23 05:58:39.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- EMC_ScaleIO http://publiclogs.emc.com/571330/4/EMC_ScaleIO/1250 : SUCCESS in 1h 37m 41s\n\nLeave a comment with \u0027run-dell-emc-scaleio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-scaleio-ci","accounts_in_message":[],"_revision_number":4},{"id":"a0140f1c33d40c4bab6ef6b09596a672889874d7","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-02 00:27:27.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"74a163026caea62a177d46baa60c4d7f08577ddb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-02 00:54:19.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/571330/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"42497b7bef0188123e10a26e5d4dc3f4071404d6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-02 01:14:37.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20952 : FAILURE in 46m 10s","accounts_in_message":[],"_revision_number":5},{"id":"4bc591718ad76372e53db2311003c59939f0d85c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-02 01:20:06.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/5/check-vote/ext-nova-zuul/bd90d20 : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":5},{"id":"0c0bb69c216dff826282bd34b6cf2f6d1c885067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-02 02:08:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/5/check/openstack-tox-pep8/db8f117/ : SUCCESS in 8m 39s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/5/check/openstack-tox-py27/7a01acd/ : SUCCESS in 11m 33s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/5/check/openstack-tox-py35/645f6be/ : SUCCESS in 12m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/5/check/build-openstack-sphinx-docs/1323e1a/html/ : SUCCESS in 6m 50s\n- tempest-full-py3 http://logs.openstack.org/30/571330/5/check/tempest-full-py3/726cb2f/ : SUCCESS in 1h 34m 37s\n- nova-tox-functional http://logs.openstack.org/30/571330/5/check/nova-tox-functional/36c8647/ : SUCCESS in 13m 42s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/5/check/nova-tox-functional-py35/9e80c3e/ : SUCCESS in 17m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/5/check/openstack-tox-lower-constraints/3af35f5/ : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":5},{"id":"58cd4fcf25ae938c17d31157198b0cfece09e359","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-02 02:14:07.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"9d33babb2a8dffad5c0a84047ef38317a5c8bb59","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-07-02 02:35:23.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/5/check/nova-out-of-tree-pvm/f122e85 : FAILURE in 2h 07m 15s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/5/check/nova-in-tree-pvm/a5973c5 : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d78d676fd30cdffb589cd3f01204aba4e10cccfa","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-07-02 06:22:22.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"089e2a4de0f4e1cd6d4b07fb1ee6a83c06d1651c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-07-02 06:27:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6648721e4f2a5114890d81610696cf82e6dcdcf5","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-04 00:36:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9b9e24f517d02ca0598ad40d5fb241bba83e3aa0","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-04 00:37:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c9caaeceb27c29635f92cf8f311fedb7d0e8776","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-04 01:03:48.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/571330/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"3e748364b9b2a635ef16b948c010000bc022d76c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-04 01:41:21.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21095 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":6},{"id":"d9530f38398953caeef4ce705de39291b963e9c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-04 02:44:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/6/check/openstack-tox-pep8/eec1784/ : SUCCESS in 12m 48s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/6/check/openstack-tox-py27/d23ae7f/ : SUCCESS in 16m 32s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/6/check/openstack-tox-py35/c78d53d/ : SUCCESS in 14m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/6/check/build-openstack-sphinx-docs/91406e2/html/ : SUCCESS in 8m 19s\n- tempest-full-py3 http://logs.openstack.org/30/571330/6/check/tempest-full-py3/750abbc/ : SUCCESS in 1h 22m 17s\n- nova-tox-functional http://logs.openstack.org/30/571330/6/check/nova-tox-functional/1bb3777/ : SUCCESS in 17m 24s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/6/check/nova-tox-functional-py35/0af96f4/ : SUCCESS in 17m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/6/check/openstack-tox-lower-constraints/e834378/ : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":6},{"id":"664d0b83d7244ebbd42d4e052af25a880ebda5cb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-04 03:20:51.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"0ce09c99d36be19286d8ac2eea6523f641ec1318","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-04 10:00:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/6/ : SUCCESS in 4h 06m 33s","accounts_in_message":[],"_revision_number":6},{"id":"15beea89e6bffd4ed951e380d015dda5932be7b0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-07-04 14:48:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-07-04/571330/6/check/tempest-dsvm-intel-nfv-xenial/d54b231 : SUCCESS in 31m 16s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-04/571330/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/bd1bfe6 : FAILURE in 1h 31m 03s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-04/571330/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/bcfc815 : SUCCESS in 57m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f232c1770698b76f583dbd1932cf22133c3d18a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-07-04 15:02:25.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"502600dff420e8ccfe767c47fa150fd229c9a686","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-04 17:06:01.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/6/check-vote/ext-nova-zuul/c9766d7 : FAILURE in 49m 35s","accounts_in_message":[],"_revision_number":6},{"id":"a8824bbfc19112d383cb4ed01df7f2839e7a2d53","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-11 22:01:35.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"bcad979598c43fda9ffff8af7418643a72389aee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-11 22:23:26.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/7/check-vote/ext-nova-zuul/f3a9e59 : FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":7},{"id":"b6928e873df16eccc1fcfd91b94f9d515f31d789","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-11 22:33:36.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/571330/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"16394fff946b68bfccf92215000876e97f5e91f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 23:28:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/7/check/openstack-tox-pep8/d0585a2/ : SUCCESS in 10m 38s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/7/check/openstack-tox-py27/a88f839/ : SUCCESS in 11m 45s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/7/check/openstack-tox-py35/072dccd/ : SUCCESS in 11m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/7/check/build-openstack-sphinx-docs/d7c0948/html/ : SUCCESS in 6m 55s\n- tempest-full-py3 http://logs.openstack.org/30/571330/7/check/tempest-full-py3/2096130/ : SUCCESS in 1h 22m 45s\n- nova-tox-functional http://logs.openstack.org/30/571330/7/check/nova-tox-functional/b7f8034/ : SUCCESS in 17m 10s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/7/check/nova-tox-functional-py35/ed9f051/ : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/7/check/openstack-tox-lower-constraints/5d5626d/ : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":7},{"id":"0982305d1a704027c9d326fb2a24f1bad64c3538","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-07-11 23:28:45.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/7/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"0e82a49f9565a4506b52a079975dc12d474ba2b8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-11 23:29:58.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21290 : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":7},{"id":"48eb5c55d895152f02392485c01dc420e327f0e7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-07-11 23:37:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-07-11/571330/7/check/tempest-dsvm-intel-nfv-xenial/3ad1779 : SUCCESS in 39m 36s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-11/571330/7/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/f00d933 : FAILURE in 1h 31m 10s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-11/571330/7/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/be781a7 : SUCCESS in 1h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7703e858a37102281582f4869d9221d960ecbf22","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-12 00:14:40.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"52d3c80c20bf51bfdbccf14445c58c4b492aab4d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-12 00:36:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/7/ : SUCCESS in 2h 34m 40s","accounts_in_message":[],"_revision_number":7},{"id":"6b9cf012ae49e5021f957d0f64321533f9d60e17","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-07-12 00:39:31.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/7/check/nova-out-of-tree-pvm/459982e : SUCCESS in 2h 37m 07s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/7/check/nova-in-tree-pvm/852c16f : FAILURE in 2h 36m 41s","accounts_in_message":[],"_revision_number":7},{"id":"3d86d782c4ee65c1fb2c1859346e014d81377fae","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-07-14 11:09:15.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b165dd6dc851f98c34db1fbba6a6ce81a9697cb5","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-17 00:02:12.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"93adf5f947a536dc1646acc9fdff9dacc258dc38","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-17 00:02:18.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"dd2ed9d4cdba7888d0bd660567895f9da92423a5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-17 00:03:53.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"3660a487f521e59d1860f68cdb0f8c6a91e3747b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-17 00:30:24.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/571330/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"fdb262a77546060c44c2015ad6bc57d56503b12e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-17 01:04:57.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21526 : FAILURE in 1h 01m 47s","accounts_in_message":[],"_revision_number":8},{"id":"aa5a64180326c06d127cc162dc7ec9208ad8e1fe","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-07-17 01:37:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-07-17/571330/8/check/tempest-dsvm-intel-nfv-xenial/6723690 : SUCCESS in 30m 54s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-17/571330/8/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e3bfc58 : FAILURE in 1h 31m 45s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-07-17/571330/8/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/8a24228 : SUCCESS in 58m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bbcbfba4a7be4540c6e590e756531223dbf1cb29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 02:03:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/8/check/openstack-tox-pep8/06c4413/ : SUCCESS in 11m 23s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/8/check/openstack-tox-py27/8b62ac0/ : SUCCESS in 11m 42s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/8/check/openstack-tox-py35/dd2f7e7/ : SUCCESS in 12m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/8/check/build-openstack-sphinx-docs/0a2862f/html/ : SUCCESS in 7m 53s\n- tempest-full-py3 http://logs.openstack.org/30/571330/8/check/tempest-full-py3/7dfa2b5/ : SUCCESS in 1h 30m 23s\n- nova-tox-functional http://logs.openstack.org/30/571330/8/check/nova-tox-functional/ce7ccdd/ : SUCCESS in 17m 41s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/8/check/nova-tox-functional-py35/ee44a8d/ : SUCCESS in 16m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/8/check/openstack-tox-lower-constraints/c83ae4d/ : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":8},{"id":"7fba9cc51dddfc8fd87d21400223c0347bbe2388","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-07-17 02:03:40.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"4c56b433cf27ddb2a516862a94df04664ac07090","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-17 03:10:05.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"ced86521f04725fcf1dfe9aa96d4b60f9d0c1047","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-17 05:05:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/8/ : SUCCESS in 3h 18m 00s","accounts_in_message":[],"_revision_number":8},{"id":"53f3b2d77fc51fc1ddf19546109fa0a44d328e6f","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2018-07-20 01:09:52.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"61742cc748b5f4d49c9a16d05549c31ee6cedbb6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-23 01:43:32.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"4f82e375a5ab5e1e3ec1f8b55219a2e1d6f69919","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-23 02:04:26.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/9/check-vote/ext-nova-zuul/8ca01a1 : FAILURE in 19m 09s","accounts_in_message":[],"_revision_number":9},{"id":"0e63de990e0f03e3b124e075451d9cef5597eaa1","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-07-23 02:10:44.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"3eda6c27f25b2078c48542ded36633da1ebdf23c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-23 02:27:58.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/571330/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"396f0e3d17606106a75be77abb2ab65b6387d704","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-23 03:06:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/9/check/openstack-tox-pep8/ae63ba9/ : SUCCESS in 8m 50s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/9/check/openstack-tox-py27/1e866e0/ : SUCCESS in 12m 09s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/9/check/openstack-tox-py35/1e36d64/ : SUCCESS in 12m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/9/check/build-openstack-sphinx-docs/b8fc6da/html/ : SUCCESS in 6m 35s\n- tempest-full-py3 http://logs.openstack.org/30/571330/9/check/tempest-full-py3/99f5cf3/ : SUCCESS in 1h 18m 34s\n- nova-tox-functional http://logs.openstack.org/30/571330/9/check/nova-tox-functional/f4d46dd/ : SUCCESS in 16m 55s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/9/check/nova-tox-functional-py35/988bd97/ : SUCCESS in 19m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/9/check/openstack-tox-lower-constraints/21177e5/ : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":9},{"id":"37e0a88cf6f37f258b0d5d0ab49d37d27600e913","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-07-23 03:06:31.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/9/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"56ea22959e32886792ab1c315b028dd287bf1dd2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-23 03:31:35.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21849 : SUCCESS in 1h 30m 29s","accounts_in_message":[],"_revision_number":9},{"id":"0cd27aa1f9315cce5ed34cfbbb7bf692e4681d7c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-23 04:21:12.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"5584e3643941bbfccd5050fbc87a39e39fc8c08b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-23 04:55:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com/nova/571330/9/ : SUCCESS in 3h 11m 09s","accounts_in_message":[],"_revision_number":9},{"id":"1c8e1f1c4376f6f357d7927865405c4c3f3e994b","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-25 01:25:59.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7e602d3d2c12505f23af3413b5333dc9f9a9d4e6","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-07-25 01:26:50.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/10/check/nova-out-of-tree-pvm/d84f1aa : FAILURE in 6s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/10/check/nova-in-tree-pvm/32b9355 : FAILURE in 5s","accounts_in_message":[],"_revision_number":10},{"id":"322768dbb9de08cb23d7c64a919bc0cf5da7e724","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-07-25 01:55:30.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/571330/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"8bebbd27b1bb0e6fb748ad066301466cd5379768","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-25 02:11:57.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/10/check-vote/ext-nova-zuul/106973f : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":10},{"id":"9af4c54d9c9403e77160a1e28dc7ba58565232a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-25 02:58:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/10/check/openstack-tox-pep8/75d40be/ : SUCCESS in 9m 32s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/10/check/openstack-tox-py27/b1497ca/ : SUCCESS in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/10/check/openstack-tox-py35/35dbe76/ : SUCCESS in 13m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/10/check/build-openstack-sphinx-docs/b35a440/html/ : SUCCESS in 10m 14s\n- tempest-full-py3 http://logs.openstack.org/30/571330/10/check/tempest-full-py3/f84dd55/ : SUCCESS in 1h 25m 40s\n- nova-tox-functional http://logs.openstack.org/30/571330/10/check/nova-tox-functional/d10d945/ : SUCCESS in 22m 12s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/10/check/nova-tox-functional-py35/901bd3a/ : SUCCESS in 18m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/10/check/openstack-tox-lower-constraints/220fd15/ : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":10},{"id":"8339922503ca3d6c02efc17166a1e626e79a7bdb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-25 03:59:27.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"0a1b9a62e6fe888676f46efa826350e21594d06b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-25 07:02:44.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-21990 : FAILURE in 5h 33m 46s","accounts_in_message":[],"_revision_number":10},{"id":"4c3d97d72b08a585d0c23b0c4ad464a29ad7bd7f","author":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"date":"2018-07-25 07:21:32.000000000","message":"Patch Set 10:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":10},{"id":"32fac1947b73e669001f4452ae42442b8b3c5950","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-25 08:47:05.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22000 : ABORTED in 1h 17m 34s","accounts_in_message":[],"_revision_number":10},{"id":"170fc4453a65825afe8d0d18e713d11e7386f197","author":{"_account_id":27076,"name":"Xiao Mei Zheng","email":"xmzheng@cn.ibm.com","username":"XiaoMeiZheng"},"date":"2018-07-25 09:58:59.000000000","message":"Patch Set 10:\n\nzvm:recheck","accounts_in_message":[],"_revision_number":10},{"id":"de36e8d704d20d8c7a0a585b5866b7b52d10ee44","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-25 16:16:47.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22009 : FAILURE in 6h 15m 59s","accounts_in_message":[],"_revision_number":10},{"id":"e75fc13074b2bdebb4d1a5d423373d96749ff696","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-31 01:04:40.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"a0352c988a491410c52073a0aa4bd52814224ad4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-31 01:49:16.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22248 : FAILURE in 44m 04s","accounts_in_message":[],"_revision_number":11},{"id":"bb892c3895b7f5a2e543707f77247fc9adb2d40d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-31 01:53:27.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/11/check-vote/ext-nova-zuul/a98c709 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":11},{"id":"3b145f90aa49e05c4c5b59967768b7acefa27808","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 02:26:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/11/check/openstack-tox-pep8/6d11f44/ : SUCCESS in 9m 24s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/11/check/openstack-tox-py27/d04c9db/ : SUCCESS in 10m 50s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/11/check/openstack-tox-py35/78ff227/ : SUCCESS in 12m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/11/check/build-openstack-sphinx-docs/2c68b14/html/ : SUCCESS in 6m 44s\n- tempest-full-py3 http://logs.openstack.org/30/571330/11/check/tempest-full-py3/19c84eb/ : FAILURE in 1h 16m 44s\n- nova-tox-functional http://logs.openstack.org/30/571330/11/check/nova-tox-functional/b0c98c5/ : SUCCESS in 18m 25s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/11/check/nova-tox-functional-py35/7bec0ea/ : SUCCESS in 17m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/11/check/openstack-tox-lower-constraints/3ec4571/ : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":11},{"id":"77a51de9ec13fd43e95bd4962300bfcfa0d819b8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-07-31 02:32:14.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"c596292f1989976349ac94bbc6be952d16fe1508","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-07-31 03:08:32.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/571330/11 : SUCCESS in 2h 03m 15s","accounts_in_message":[],"_revision_number":11},{"id":"3d649907a52326817e1d5dfb423c7ec0a43735c8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-07-31 03:20:00.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/11/check-vote/ext-nova-zuul/d9d3451 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":11},{"id":"68e598a20d4a8ab7f24271007892c5dc9244766d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-07-31 03:28:07.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22251 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":11},{"id":"16718b1ccc903affa85dd7199efda45df7895f98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-31 04:36:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/11/check/openstack-tox-pep8/be22645/ : SUCCESS in 9m 26s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/11/check/openstack-tox-py27/0d3422c/ : SUCCESS in 12m 00s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/11/check/openstack-tox-py35/b59739f/ : SUCCESS in 13m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/11/check/build-openstack-sphinx-docs/df5a61c/html/ : SUCCESS in 6m 41s\n- tempest-full-py3 http://logs.openstack.org/30/571330/11/check/tempest-full-py3/1c229a7/ : SUCCESS in 2h 02m 34s\n- nova-tox-functional http://logs.openstack.org/30/571330/11/check/nova-tox-functional/d424969/ : SUCCESS in 19m 14s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/11/check/nova-tox-functional-py35/fe01aaf/ : SUCCESS in 17m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/11/check/openstack-tox-lower-constraints/c4c4018/ : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":11},{"id":"a4a396d147b3e96883baa98b55958d2935a5033e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-07-31 04:37:01.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/11/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"13836c2145fd7ea40ce81c32e19954d61ac7b141","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-07-31 05:29:49.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"b0771dd95079a5fda4c46d981e0572e2e968e1ae","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-08-01 14:43:46.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"f04873dddf2ed09934740c3868e2bc1837a141e9","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2018-08-01 22:51:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"00ad76e7e35543e1cda4397ab6831f83de4baca4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-02 23:15:53.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-02/571330/11/check/tempest-dsvm-intel-nfv-xenial/f14b62b : EXCEPTION (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-02/571330/11/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/455dceb : EXCEPTION (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-02/571330/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/40dde9f : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e565b7f2cdc5ad417e6791860c7b30edee829b97","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-08-03 07:10:13.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"4d1fbc23c62a5c00092d22afc93cf9932da78151","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-08-05 00:16:57.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"4054c09e28b5708d1d2e23581e77efddd9016ca9","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-08-06 03:17:11.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"9b2e63d77e9d391cd7fc2496b455dafa2ea57db3","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-08-13 12:10:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3ecab7e01c363943a062cc26efe1308ae5c8f44a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-13 12:30:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/12/check/openstack-tox-pep8/f8b3798/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/12/check/openstack-tox-py27/9e2dfe2/ : SUCCESS in 12m 06s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/12/check/openstack-tox-py35/6abcc3b/ : SUCCESS in 12m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/12/check/build-openstack-sphinx-docs/d07642a/html/ : SUCCESS in 7m 27s\n- nova-tox-functional http://logs.openstack.org/30/571330/12/check/nova-tox-functional/27b685b/ : SUCCESS in 17m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/12/check/nova-tox-functional-py35/f53000a/ : SUCCESS in 15m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/12/check/openstack-tox-lower-constraints/fa6d160/ : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":12},{"id":"f8e2ea20738de7b5556ba11e588c58e5f4cf750e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-08-13 12:31:18.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/12/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":12},{"id":"4a3c914ec5602c2bb55ffaae8570a20abc7af6f9","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-13 12:58:44.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/571330/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"264bb3ba644748ef9b2f39d31a877ae9b1d22c4c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-13 12:59:05.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/12/check-vote/ext-nova-zuul/077d669 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":12},{"id":"345460a8d7ea9b8a13ab7e3a2f9b7e85b8ebd988","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-13 13:53:15.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22674 : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":12},{"id":"21bdcf1d2071886c19efa3c13e7a6f1fa86e18c2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-13 14:32:04.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/571330/12 : FAILURE in 2h 21m 23s","accounts_in_message":[],"_revision_number":12},{"id":"489681a160c38b3d16a210a8eea9b3715b99c6be","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2018-08-23 17:19:44.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"9299ff022a0f3f26e7d6123358f738eb6655896e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-08-26 22:05:50.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"e378390216ffc45f066d2811de696781cf7d1596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-26 22:30:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/13/check/openstack-tox-pep8/235df63/ : SUCCESS in 9m 39s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/13/check/openstack-tox-py27/4245f30/ : SUCCESS in 12m 23s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/13/check/openstack-tox-py35/2acaf68/ : SUCCESS in 11m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/13/check/build-openstack-sphinx-docs/b45e8be/html/ : SUCCESS in 7m 04s\n- nova-tox-functional http://logs.openstack.org/30/571330/13/check/nova-tox-functional/efe316a/ : SUCCESS in 17m 41s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/13/check/nova-tox-functional-py35/9a7666a/ : SUCCESS in 16m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/13/check/openstack-tox-lower-constraints/84bb601/ : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":13},{"id":"c52b5c75ecdc0381fe2437539d1dda235aedf4d4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-26 22:32:44.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/571330/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"8cf72a59d4cecdb158f630b67576958bcb65d138","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-26 22:59:05.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/13/check-vote/ext-nova-zuul/af9437f : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":13},{"id":"4a2e527f5d234513ec59fc65125032e333c88815","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-26 23:24:10.000000000","message":"Patch Set 13:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/13/check/nova-out-of-tree-pvm/e404706 : SUCCESS in 1h 17m 57s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/13/check/nova-in-tree-pvm/baa84f0 : FAILURE in 39m 51s","accounts_in_message":[],"_revision_number":13},{"id":"419b8e1abf972fe5cd1b11b5f6c1f5bda94d7109","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-08-26 23:27:18.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-13 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":13},{"id":"61e4c4e9bce1b0bbd3114a4e6380b0f55e17d74b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-26 23:42:39.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-26/571330/13/check/tempest-dsvm-intel-nfv-xenial/46d6f38 : SUCCESS in 50m 19s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-26/571330/13/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/0ecca61 : FAILURE in 1h 32m 25s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-26/571330/13/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6961638 : SUCCESS in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c4c5907c9c28501fa4e4e2bbe3bf87f209822e65","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-27 00:10:13.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/571330/13 : SUCCESS in 2h 03m 51s","accounts_in_message":[],"_revision_number":13},{"id":"5838bb552f0e22f12a28edb4018efcc828f2b29c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-28 04:07:04.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-23191 : SUCCESS in 2h 17m 48s","accounts_in_message":[],"_revision_number":13},{"id":"de7c63114b2d4e16193f7c98fdd2cf6d7a0a14f2","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2018-08-29 03:36:28.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"9fd664fc5aad7383ae7ca5f217ac0536ef456aea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-09-03 11:38:19.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"a063244055d97f25b3ef557884bca5cfd27d76b3","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-09-18 02:09:40.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"ec12ae48ed5d44253b72c761f44398b1bcc30991","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-18 02:33:21.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/14/check/openstack-tox-pep8/f2c9f05/ : SUCCESS in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/14/check/openstack-tox-py27/362ed17/ : SUCCESS in 12m 09s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/14/check/openstack-tox-py35/16510fb/ : SUCCESS in 12m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/14/check/build-openstack-sphinx-docs/3abba75/html/ : SUCCESS in 7m 05s\n- openstack-tox-cover http://logs.openstack.org/30/571330/14/check/openstack-tox-cover/bb402a9/cover/ : SUCCESS in 14m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/14/check/openstack-tox-lower-constraints/90752c0/ : SUCCESS in 11m 29s\n- openstack-tox-py36 http://logs.openstack.org/30/571330/14/check/openstack-tox-py36/a8c2062/ : SUCCESS in 15m 12s\n- openstack-tox-docs http://logs.openstack.org/30/571330/14/check/openstack-tox-docs/62831dc/html/ : SUCCESS in 6m 17s\n- nova-tox-functional http://logs.openstack.org/30/571330/14/check/nova-tox-functional/3dbd577/ : SUCCESS in 16m 19s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/14/check/nova-tox-functional-py35/9685953/ : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":14},{"id":"8714639ecab54345dfc9301b32d60d37c7418528","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-09-18 02:33:43.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/571330/14/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":14},{"id":"07f8cd74c5f0ec4b84e0da4431d0970c544cdd93","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-09-18 02:36:40.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/571330/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"ad65447ddb2f75df6227ff3e3aa0b945b0e845ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-09-18 02:40:20.000000000","message":"Patch Set 14:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-30-571330-14 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":14},{"id":"ba12540e9a213dd654887a8eb51b7a9d2fa9a1d8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2018-09-18 03:44:10.000000000","message":"Patch Set 14: Code-Review+1\n\nLooks good, but man the size of test_drive is a little nuts.","accounts_in_message":[],"_revision_number":14},{"id":"17bc7fb14e24e6260ec1e811aa2e5f9c5a6522d3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-09-18 04:19:30.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/571330/14 : SUCCESS in 2h 03m 56s","accounts_in_message":[],"_revision_number":14},{"id":"b8c2d719e67ffc9a1a1b88b1aac37c854878d845","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-09-18 05:54:02.000000000","message":"Patch Set 14:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/30/571330/14/check/nova-out-of-tree-pvm/8a6be2a : SUCCESS in 3h 02m 06s\n- nova-in-tree-pvm http://184.172.12.213/30/571330/14/check/nova-in-tree-pvm/7846210 : FAILURE in 2h 46m 41s","accounts_in_message":[],"_revision_number":14},{"id":"35816c404eaab3a38ecb590bef2ab38ac13c77ed","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-09-18 08:28:06.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-23746 : SUCCESS in 2h 49m 27s","accounts_in_message":[],"_revision_number":14},{"id":"ae50b01c33de633a0e69f89d9e7f1659b44e575f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-09-18 21:45:58.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/30/571330/14/check-vote/ext-nova-zuul/1c9cb9b : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":14},{"id":"c629ebff4df31a9d2c5fa94a0fd530cfb48714be","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2018-09-21 09:55:44.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"c4bec53aeef2bfcc767cc25be4f6ee6ea621c7cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 12:09:24.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"9f4a9af35c66bd4e68708b9720de08a193b10770","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 13:59:07.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/30/571330/14/gate/openstack-tox-pep8/a4e15b7/ : SUCCESS in 9m 10s\n- openstack-tox-py27 http://logs.openstack.org/30/571330/14/gate/openstack-tox-py27/30976f4/ : SUCCESS in 15m 39s\n- openstack-tox-py35 http://logs.openstack.org/30/571330/14/gate/openstack-tox-py35/1550559/ : SUCCESS in 15m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/30/571330/14/gate/build-openstack-sphinx-docs/0964b30/html/ : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/571330/14/gate/openstack-tox-lower-constraints/ba9d739/ : SUCCESS in 14m 12s\n- openstack-tox-py36 http://logs.openstack.org/30/571330/14/gate/openstack-tox-py36/352384b/ : SUCCESS in 13m 06s\n- openstack-tox-docs http://logs.openstack.org/30/571330/14/gate/openstack-tox-docs/91718d0/html/ : SUCCESS in 8m 03s\n- nova-tox-functional http://logs.openstack.org/30/571330/14/gate/nova-tox-functional/38eb118/ : SUCCESS in 18m 16s\n- nova-tox-functional-py35 http://logs.openstack.org/30/571330/14/gate/nova-tox-functional-py35/3799c65/ : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":14},{"id":"58d12f60a48a1000892c71be3da850795e94a522","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 13:59:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"56deff48a616562b439d68292a14e8eebdd9ee11","revisions":{"09b1cd25f932789d5ef13495c4bd334286f5ea00":{"kind":"REWORK","_number":1,"created":"2018-05-30 23:20:09.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/1"}}},"commit":{"parents":[{"commit":"5070d67d7259b79bd5417eea4ff7e520006a6010","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5070d67d7259b79bd5417eea4ff7e520006a6010"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 23:19:34.000000000","tz":540},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09b1cd25f932789d5ef13495c4bd334286f5ea00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09b1cd25f932789d5ef13495c4bd334286f5ea00"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5070d67d7259b79bd5417eea4ff7e520006a6010","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f3056d90d6098258592c5f82da89299bd1f11a49":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-06-03 22:45:26.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/2"}}},"commit":{"parents":[{"commit":"aa35cfbca4f165d0df72a950bf43eed93e1185fe","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa35cfbca4f165d0df72a950bf43eed93e1185fe"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-06-03 22:45:26.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3056d90d6098258592c5f82da89299bd1f11a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3056d90d6098258592c5f82da89299bd1f11a49"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aa35cfbca4f165d0df72a950bf43eed93e1185fe","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3e95e579aefc8c89758ecfb5ca19b2b59c6bfea8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-06-07 21:37:38.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/3"}}},"commit":{"parents":[{"commit":"e69292736bd79517b755cafdf83e1ef0e1e4345b","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e69292736bd79517b755cafdf83e1ef0e1e4345b"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-06-07 21:37:38.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e95e579aefc8c89758ecfb5ca19b2b59c6bfea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3e95e579aefc8c89758ecfb5ca19b2b59c6bfea8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e69292736bd79517b755cafdf83e1ef0e1e4345b","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3a8494966f9b1273a9b4d308d251b808504c5d65":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-06-22 04:31:06.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/4"}}},"commit":{"parents":[{"commit":"fef34cb92b4241b1f7c87d0cc2c2b447d3666757","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fef34cb92b4241b1f7c87d0cc2c2b447d3666757"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-06-22 04:31:06.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a8494966f9b1273a9b4d308d251b808504c5d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a8494966f9b1273a9b4d308d251b808504c5d65"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fef34cb92b4241b1f7c87d0cc2c2b447d3666757","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"781c6c02f685272d1880a26e15dbbacbfb1584a0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-07-02 00:27:27.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/5"}}},"commit":{"parents":[{"commit":"5ee87f25ca42794c71c337c70a691530ba4640bb","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ee87f25ca42794c71c337c70a691530ba4640bb"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-02 00:27:27.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/781c6c02f685272d1880a26e15dbbacbfb1584a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/781c6c02f685272d1880a26e15dbbacbfb1584a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5ee87f25ca42794c71c337c70a691530ba4640bb","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d318ca2fb3bbc8adc878f4c83e57b55247f7fc61":{"kind":"REWORK","_number":6,"created":"2018-07-04 00:36:53.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/6"}}},"commit":{"parents":[{"commit":"c8d7613f6bcc6a269cad24690851cdb8b9433a3b","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c8d7613f6bcc6a269cad24690851cdb8b9433a3b"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-04 00:36:28.000000000","tz":540},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d318ca2fb3bbc8adc878f4c83e57b55247f7fc61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d318ca2fb3bbc8adc878f4c83e57b55247f7fc61"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c8d7613f6bcc6a269cad24690851cdb8b9433a3b","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ae912950adadb658368cc8a1d7e26468162308e8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2018-07-11 22:01:35.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/7"}}},"commit":{"parents":[{"commit":"24c4954338e06dd615782b68a131ccccc968e929","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24c4954338e06dd615782b68a131ccccc968e929"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-11 22:01:35.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae912950adadb658368cc8a1d7e26468162308e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae912950adadb658368cc8a1d7e26468162308e8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"24c4954338e06dd615782b68a131ccccc968e929","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":8,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c6751ac817eca6cc3496ae62926ebced68dc3a71":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-07-17 00:02:18.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/8"}}},"commit":{"parents":[{"commit":"702d24b558ac6e0b9387fe81d83a4a25810ba644","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/702d24b558ac6e0b9387fe81d83a4a25810ba644"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-17 00:02:18.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6751ac817eca6cc3496ae62926ebced68dc3a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6751ac817eca6cc3496ae62926ebced68dc3a71"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"702d24b558ac6e0b9387fe81d83a4a25810ba644","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":10,"change_status":"MERGED"}],"branch":"refs/heads/master"},"014e2c57bd9fd7aad8d5b86406db58558549779e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-07-23 01:43:32.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/9"}}},"commit":{"parents":[{"commit":"0078074955ba071871d4834aedd9a1173399076b","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0078074955ba071871d4834aedd9a1173399076b"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-23 01:43:32.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/014e2c57bd9fd7aad8d5b86406db58558549779e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/014e2c57bd9fd7aad8d5b86406db58558549779e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0078074955ba071871d4834aedd9a1173399076b","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"df1f3443f4376eb7b05623cd9946ddea529d4516":{"kind":"REWORK","_number":10,"created":"2018-07-25 01:25:59.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/10"}}},"commit":{"parents":[{"commit":"3a5e91591f26384154e2819bb5b3ae770eb17ee5","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a5e91591f26384154e2819bb5b3ae770eb17ee5"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-25 01:25:10.000000000","tz":540},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df1f3443f4376eb7b05623cd9946ddea529d4516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df1f3443f4376eb7b05623cd9946ddea529d4516"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3a5e91591f26384154e2819bb5b3ae770eb17ee5","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":12,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9159879d46db9599ea80b6dd78513f28672c768e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2018-07-31 01:04:40.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/11"}}},"commit":{"parents":[{"commit":"9ddcd2961b26d84925dbe7924a98a45208b26316","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ddcd2961b26d84925dbe7924a98a45208b26316"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-07-31 01:04:40.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9159879d46db9599ea80b6dd78513f28672c768e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9159879d46db9599ea80b6dd78513f28672c768e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ddcd2961b26d84925dbe7924a98a45208b26316","is_merged_in_target_branch":false,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e1a01092da6bc6e23ae5cc23850dac0b35b7ce3e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2018-08-13 12:10:05.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/12"}}},"commit":{"parents":[{"commit":"2baece04af0ab18fa604fa4d72c585a442df47cd","subject":"Remove mox in libvirt/test_driver.py (5)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2baece04af0ab18fa604fa4d72c585a442df47cd"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-08-13 12:08:40.000000000","tz":540},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal-stein\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1a01092da6bc6e23ae5cc23850dac0b35b7ce3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1a01092da6bc6e23ae5cc23850dac0b35b7ce3e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2baece04af0ab18fa604fa4d72c585a442df47cd","is_merged_in_target_branch":true,"change_id":"I530da7aed52d0cbaf0a4451aaf1442c12ed69642","change_number":570842,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2da1b1e5495992f31d5542dd3566a9549e67febd":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2018-08-26 22:05:50.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/13"}}},"commit":{"parents":[{"commit":"8ee44744e69a1f82c1bf2feb8b884440794f2b6c","subject":"Merge \"Batch results per cell when doing cross-cell listing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ee44744e69a1f82c1bf2feb8b884440794f2b6c"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-08-26 22:05:50.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal-stein\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da1b1e5495992f31d5542dd3566a9549e67febd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da1b1e5495992f31d5542dd3566a9549e67febd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8ee44744e69a1f82c1bf2feb8b884440794f2b6c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"56deff48a616562b439d68292a14e8eebdd9ee11":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2018-09-18 02:09:40.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/30/571330/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/30/571330/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/30/571330/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/30/571330/14"}}},"commit":{"parents":[{"commit":"82270cc261f6c1d9d2cc386f1fb445dd66023f75","subject":"Merge \"Fix docs and add functional test for AggregateMultiTenancyIsolation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82270cc261f6c1d9d2cc386f1fb445dd66023f75"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-05-30 20:37:30.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2018-09-18 02:09:40.000000000","tz":0},"subject":"Remove mox in libvirt/test_driver.py (6)","message":"Remove mox in libvirt/test_driver.py (6)\n\nReplace mox with mock or stub_out in the following methods of\nthe LibvirtDriverTestCase\nin nova/tests/unit/virt/libvirt/test_driver.py.\n\n* _test_migrate_disk_and_power_off_resize_check\n* test_migrate_disk_and_power_off_lvm\n* test_migrate_disk_and_power_off_resize_cannot_ssh\n* test_wait_for_running\n* _test_finish_revert_migration\n* test_cleanup_failed_migration\n* test_confirm_migration\n* test_get_instance_disk_info_exception\n* _test_attach_interface\n* test_detach_interface_device_with_same_mac_address\n\nSubsequent patches will replace mox in other test cases.\n\nChange-Id: I0cd5fc54d981acbf6a63ae74e272f83f24246c70\nImplements: blueprint mox-removal-stein\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56deff48a616562b439d68292a14e8eebdd9ee11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56deff48a616562b439d68292a14e8eebdd9ee11"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"82270cc261f6c1d9d2cc386f1fb445dd66023f75","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
