)]}'
{"id":"openstack%2Fnova~577385","triplet_id":"openstack%2Fnova~master~Ia39402a045ffb1943463b5741655d84071613e8c","project":"openstack/nova","branch":"master","topic":"Update-default-for-rng_dev_path","hashtags":[],"change_id":"Ia39402a045ffb1943463b5741655d84071613e8c","subject":"conf: libvirt: Make `/dev/urandom` the default for \u0027rng_dev_path\u0027","status":"MERGED","created":"2018-06-22 10:17:24.000000000","updated":"2019-04-05 13:31:44.000000000","submitted":"2018-06-29 07:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"577385-1530256729545-0acd0a43","meta_rev_id":"e412ce4b80445661b4bbc8fab49c0931d56a80f8","_number":577385,"virtual_id_number":577385,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"date":"2018-06-26 19:34:37.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-26 19:48:34.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-27 04:26:38.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-26 17:45:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2018-06-26 19:32:29.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"date":"2018-06-26 18:38:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-06-29 07:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-26 18:27:27.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-06-26 19:11:49.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"date":"2018-06-27 06:42:24.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2018-06-26 21:16:34.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"all":[{"value":1,"date":"2018-06-26 16:22:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2018-06-27 15:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-27 06:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-27 06:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-06-28 22:16:21.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-06-28 22:16:21.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-26 16:22:50.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2018-06-26 17:45:52.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 18:27:27.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 18:38:54.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-06-26 19:11:49.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 19:32:29.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 19:34:37.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 19:48:34.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-26 21:16:34.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-27 04:26:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-06-27 06:16:39.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2018-06-27 06:42:24.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-06-27 06:43:05.000000000","updated_by":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"reviewer":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"state":"REVIEWER"},{"updated":"2018-06-27 15:00:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2018-06-28 22:16:21.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-06-29 07:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e20b2654510afdaa455c04770d714ae921dea452","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-22 10:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a52629c8746c4dc516b14671a08b732af5d788b8","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-22 10:18:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"659cad5b42a845ab1a470ef5d269f8de474b6ae5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-22 11:19:25.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/85/577385/2/check-vote/ext-nova-zuul/9eeb3f9 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"75e848f7fa8d19c0742dd8f88f62903fa9de96bf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-06-22 11:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/2/check/tempest-dsvm-full-xenial/8c0c15a/ : SUCCESS in 1h 05m 43s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/2/check/grenade-dsvm-xenial/bcd2875/ : SUCCESS in 47m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87067b012ce44f08a886716c4e8a2abbb1cb52b0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-06-22 11:47:59.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/577385/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/62df9c7 : FAILURE in 1h 26m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c7bdf85b61e90df89563991113fa4ec4843e3baa","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-22 11:52:11.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20557 : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b9aa94deb0e454f7c654f53a031cc9352903d232","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-06-22 12:13:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/85/577385/2/check/dsvm-tempest-neutron-network/6a63ec6 : SUCCESS in 1h 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"179920db7f7d31f650d445548c9f4077c6a752aa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-06-22 12:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-06-22/577385/2/check/tempest-dsvm-intel-nfv-xenial/ccd4bf4 : SUCCESS in 30m 34s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/577385/2/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/ff75756 : FAILURE in 1h 30m 50s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/577385/2/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/d0d9620 : SUCCESS in 57m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"667d8713882ca7b92e0e7610ceaaa9bccdb691da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 12:25:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/577385/2/check/openstack-tox-pep8/bd758e0/ : SUCCESS in 10m 07s\n- openstack-tox-py27 http://logs.openstack.org/85/577385/2/check/openstack-tox-py27/adeb3e4/ : SUCCESS in 13m 17s\n- openstack-tox-py35 http://logs.openstack.org/85/577385/2/check/openstack-tox-py35/d0818fe/ : SUCCESS in 14m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/577385/2/check/build-openstack-sphinx-docs/e2df47c/html/ : SUCCESS in 8m 34s\n- tempest-full http://logs.openstack.org/85/577385/2/check/tempest-full/aca9654/ : SUCCESS in 1h 37m 00s\n- neutron-grenade http://logs.openstack.org/85/577385/2/check/neutron-grenade/997b093/ : SUCCESS in 46m 17s\n- tempest-full-py3 http://logs.openstack.org/85/577385/2/check/tempest-full-py3/1200287/ : SUCCESS in 1h 25m 32s\n- neutron-grenade-multinode http://logs.openstack.org/85/577385/2/check/neutron-grenade-multinode/7d2e453/ : SUCCESS in 55m 41s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/577385/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/fa491d4/ : SUCCESS in 56m 27s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/577385/2/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/5055c03/ : SUCCESS in 1h 04m 57s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/85/577385/2/check/neutron-tempest-multinode-full/4b60fcf/ : SUCCESS in 1h 37m 44s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/577385/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/84fd2bd/ : SUCCESS in 38m 28s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/85/577385/2/check/nova-cells-v1/cf77716/ : SUCCESS in 56m 33s\n- nova-live-migration http://logs.openstack.org/85/577385/2/check/nova-live-migration/430b6b0/ : SUCCESS in 38m 06s\n- nova-multiattach http://logs.openstack.org/85/577385/2/check/nova-multiattach/e514a7f/ : SUCCESS in 50m 34s\n- nova-next http://logs.openstack.org/85/577385/2/check/nova-next/95ace6b/ : SUCCESS in 1h 49m 10s\n- nova-tox-functional http://logs.openstack.org/85/577385/2/check/nova-tox-functional/025efd4/ : SUCCESS in 18m 10s\n- nova-tox-functional-py35 http://logs.openstack.org/85/577385/2/check/nova-tox-functional-py35/f3eab38/ : SUCCESS in 14m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/577385/2/check/openstack-tox-lower-constraints/cfdc112/ : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9d493c25d342aa53c98e839bb2607f71d3550223","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-06-22 15:07:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/85/577385/2/check-nova/Nova-ML2-Sriov/a6cfc85 : SUCCESS in 56m 28s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/85/577385/2/check-nova/Nova-MACVTAP-ML2-Sriov/1c56c5d : SUCCESS in 41m 40s (non-voting)\n- NVMe http://13.74.249.42/85/577385/2/check-nova/NVMe/0e337d3 : SUCCESS in 46m 14s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"ee8524c69cf789789fb6e6647c833c05adaaf782","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-22 15:51:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa5b46b92b16896278596a0ab135b7493e3792b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-22 16:08:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5432b3dfef381758c99adbd2b9d15fe03614cb1","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-06-22 17:11:52.000000000","message":"Patch Set 3: Code-Review-1\n\nThe help text looks fine. \n\nShould we maybe set the default value to \u0027/dev/urandom\u0027?\n\nAccording to \"https://libvirt.org/formatdomain.html#elementsRng\" if no filename is specified (which seems to be the case by default currently) then qemu will use \"/dev/random\" by default.","accounts_in_message":[],"_revision_number":3},{"id":"83eb12af4e6051ea4e5f472f7ba09d422c24f528","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 17:55:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/577385/3/check/openstack-tox-pep8/d62bcca/ : SUCCESS in 9m 56s\n- openstack-tox-py27 http://logs.openstack.org/85/577385/3/check/openstack-tox-py27/87ab568/ : SUCCESS in 11m 38s\n- openstack-tox-py35 http://logs.openstack.org/85/577385/3/check/openstack-tox-py35/ac66278/ : SUCCESS in 13m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/577385/3/check/build-openstack-sphinx-docs/0bfd9ea/html/ : SUCCESS in 7m 11s\n- tempest-full http://logs.openstack.org/85/577385/3/check/tempest-full/30ef0e3/ : SUCCESS in 1h 35m 21s\n- neutron-grenade http://logs.openstack.org/85/577385/3/check/neutron-grenade/ff002f3/ : SUCCESS in 56m 21s\n- tempest-full-py3 http://logs.openstack.org/85/577385/3/check/tempest-full-py3/e1021bd/ : SUCCESS in 1h 13m 38s\n- neutron-grenade-multinode http://logs.openstack.org/85/577385/3/check/neutron-grenade-multinode/cc6c945/ : SUCCESS in 1h 04m 10s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/577385/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/207487c/ : SUCCESS in 1h 04m 38s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/577385/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/ef0a2cc/ : SUCCESS in 1h 06m 27s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/85/577385/3/check/neutron-tempest-multinode-full/adf2e94/ : SUCCESS in 1h 53m 54s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/577385/3/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ab7e8a0/ : SUCCESS in 49m 08s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/85/577385/3/check/nova-cells-v1/37145ab/ : SUCCESS in 53m 24s\n- nova-live-migration http://logs.openstack.org/85/577385/3/check/nova-live-migration/36a1548/ : SUCCESS in 40m 51s\n- nova-multiattach http://logs.openstack.org/85/577385/3/check/nova-multiattach/7214a8e/ : SUCCESS in 50m 00s\n- nova-next http://logs.openstack.org/85/577385/3/check/nova-next/c0a8a8e/ : SUCCESS in 1h 33m 19s\n- nova-tox-functional http://logs.openstack.org/85/577385/3/check/nova-tox-functional/692d181/ : SUCCESS in 16m 48s\n- nova-tox-functional-py35 http://logs.openstack.org/85/577385/3/check/nova-tox-functional-py35/9ffd226/ : SUCCESS in 15m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/577385/3/check/openstack-tox-lower-constraints/79f9eec/ : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":3},{"id":"37c37d1b22408afae6ec881794c79a29cd443934","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-06-22 18:46:23.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/3/check/tempest-dsvm-full-xenial/867e593/ : SUCCESS in 1h 02m 56s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/3/check/grenade-dsvm-xenial/2c70fcd/ : SUCCESS in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba7abd6e3c739b6781779d0bd41828b404419b1f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-22 19:00:21.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/85/577385/3/check-vote/ext-nova-zuul/c52b956 : FAILURE in 1h 09m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c534cefdb253357a35a6877a7365b311d7ec1631","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-06-22 19:21:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-06-22/577385/3/check/tempest-dsvm-intel-nfv-xenial/dc30ed8 : SUCCESS in 31m 31s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/577385/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/45ff42f : FAILURE in 1h 31m 36s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-22/577385/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9a3c440 : SUCCESS in 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ac6f59acc2e2f4334906e821330b9d600c7f266","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-22 19:28:35.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/577385/3/check/nova-out-of-tree-pvm/0ce05b8 : FAILURE in 2h 03m 08s\n- nova-in-tree-pvm http://184.172.12.213/85/577385/3/check/nova-in-tree-pvm/749bca1 : FAILURE in 2h 10m 33s","accounts_in_message":[],"_revision_number":3},{"id":"333f94cbd9bc4e60899fa4b1ed9a2b8fc4812eaa","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-06-22 19:37:19.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/577385/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ca90321 : SUCCESS in 1h 44m 46s","accounts_in_message":[],"_revision_number":3},{"id":"bcf54838e7ce9aab3a5804e3a86714c64d507c32","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-06-22 20:19:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/85/577385/3/check/dsvm-tempest-neutron-network/3981b35 : SUCCESS in 2h 40m 03s","accounts_in_message":[],"_revision_number":3},{"id":"c96cdf2ad131ee0d2deffcaff4007ef3502d6987","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-22 22:42:38.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20585 : FAILURE in 2h 29m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a592cbff70d56809d132d57d07c9d0b6889df6a8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-23 08:13:54.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/577385/3/ : FAILURE in 2h 48m 21s","accounts_in_message":[],"_revision_number":3},{"id":"976d138bf419ca12ffeb2136352437af384cbdd4","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-25 11:27:28.000000000","message":"Patch Set 3:\n\n\u003e The help text looks fine.\n \u003e \n \u003e Should we maybe set the default value to \u0027/dev/urandom\u0027?\n \u003e \n \u003e According to \"https://libvirt.org/formatdomain.html#elementsRng\" if\n \u003e no filename is specified (which seems to be the case by default\n \u003e currently) then qemu will use \"/dev/random\" by default.\n\nYes, Nova can set default to /dev/urandom.  (Also, most other projects are overriding the QEMU default to /dev/urandom anyway; e.g. the libguestfs project: https://github.com/libguestfs/libguestfs/blob/master/lib/launch-direct.c#L592)\n\nAnd QEMU upstream itself should default to /dev/urandom; I raised it on their upstream mailing list.","accounts_in_message":[],"_revision_number":3},{"id":"6093464c29041849133c5017c28ee0788aa484e3","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-25 12:09:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"789d74ce856a1f8bdcb03399cadb3d255a0571bd","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-06-25 12:37:07.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/577385/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1ecf264bb4fc5f184cc96465fc53ca1c9468d866","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-25 12:59:19.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/85/577385/4/check-vote/ext-nova-zuul/132e6cf : FAILURE in 48m 46s","accounts_in_message":[],"_revision_number":4},{"id":"e3f59e182a8ddccb08bfebf0d2b9f6b52d7bc990","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-06-25 13:10:13.000000000","message":"Patch Set 4:\n\nDon\u0027t you need to add a reno note when you change the default behavior of an option?","accounts_in_message":[],"_revision_number":4},{"id":"d96b77f317870f42f50709948640d693f1a19623","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-25 13:14:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1e2bd2929027bf8cf392595cf05e85fc46b33b56","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-25 13:14:57.000000000","message":"Patch Set 4: Code-Review-1\n\n\u003e Don\u0027t you need to add a reno note when you change the default\n \u003e behavior of an option?\n\nOh, good point. There should probably be a small reno for this, yes","accounts_in_message":[],"_revision_number":4},{"id":"586c89d61dde137815c6a82b305e8cd99552b293","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-06-25 13:22:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/4/check/tempest-dsvm-full-xenial/0c1e151/ : SUCCESS in 1h 12m 12s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/4/check/grenade-dsvm-xenial/da6b1e1/ : SUCCESS in 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fda2c847553fa3547c1cc22008e3a4b2b290a964","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-06-25 13:23:03.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/577385/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c90620c : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"5b88c95adcd20f4b37dbb49caf2c3c460817c553","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-25 13:45:53.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/577385/4/check/nova-out-of-tree-pvm/78f5565 : FAILURE in 1h 12m 35s\n- nova-in-tree-pvm http://184.172.12.213/85/577385/4/check/nova-in-tree-pvm/2801f58 : FAILURE in 1h 35m 51s","accounts_in_message":[],"_revision_number":4},{"id":"f35ee2d22f4eac789d5c8e797af77ae078fa8154","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-25 13:49:04.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20624 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":4},{"id":"a87afa48df7ef0e27eef3079e749284b04781187","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-06-25 14:06:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/85/577385/4/check-nova/Nova-ML2-Sriov/4d5c8cf : SUCCESS in 57m 20s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/85/577385/4/check-nova/Nova-MACVTAP-ML2-Sriov/cdb19c9 : SUCCESS in 47m 00s (non-voting)\n- NVMe http://13.74.249.42/85/577385/4/check-nova/NVMe/520d276 : SUCCESS in 51m 14s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"2dc6f79d3753cb963fe087494b70214498c21afd","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-06-25 14:15:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/85/577385/4/check/dsvm-tempest-neutron-network/8b7fc32 : SUCCESS in 2h 04m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ed45f408e387d529042df88901656d99ce84d47e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 14:15:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/85/577385/4/check/openstack-tox-pep8/1a5c538/ : SUCCESS in 10m 24s\n- openstack-tox-py27 http://logs.openstack.org/85/577385/4/check/openstack-tox-py27/0d1c50d/ : FAILURE in 12m 30s\n- openstack-tox-py35 http://logs.openstack.org/85/577385/4/check/openstack-tox-py35/334bc39/ : FAILURE in 13m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/577385/4/check/build-openstack-sphinx-docs/cc96a0f/html/ : SUCCESS in 7m 30s\n- tempest-full http://logs.openstack.org/85/577385/4/check/tempest-full/f518211/ : SUCCESS in 1h 42m 34s\n- neutron-grenade http://logs.openstack.org/85/577385/4/check/neutron-grenade/31b2e73/ : SUCCESS in 53m 21s\n- tempest-full-py3 http://logs.openstack.org/85/577385/4/check/tempest-full-py3/b6da8a2/ : SUCCESS in 1h 34m 37s\n- neutron-grenade-multinode http://logs.openstack.org/85/577385/4/check/neutron-grenade-multinode/eeee524/ : SUCCESS in 57m 45s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/577385/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/ed5608f/ : SUCCESS in 1h 05m 15s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/577385/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/40fec98/ : SUCCESS in 1h 09m 07s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/85/577385/4/check/neutron-tempest-multinode-full/dd4aae3/ : SUCCESS in 1h 52m 43s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/577385/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0532980/ : SUCCESS in 42m 45s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/85/577385/4/check/nova-cells-v1/f38c220/ : SUCCESS in 52m 00s\n- nova-live-migration http://logs.openstack.org/85/577385/4/check/nova-live-migration/17c2494/ : SUCCESS in 40m 52s\n- nova-multiattach http://logs.openstack.org/85/577385/4/check/nova-multiattach/9ff68ca/ : SUCCESS in 50m 53s\n- nova-next http://logs.openstack.org/85/577385/4/check/nova-next/a7ddaa7/ : SUCCESS in 1h 35m 30s\n- nova-tox-functional http://logs.openstack.org/85/577385/4/check/nova-tox-functional/de86c60/ : SUCCESS in 15m 40s\n- nova-tox-functional-py35 http://logs.openstack.org/85/577385/4/check/nova-tox-functional-py35/f639765/ : SUCCESS in 17m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/577385/4/check/openstack-tox-lower-constraints/a7e6262/ : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":4},{"id":"97447a0b0c2db3b5b0e2bab2b41ef6fa1e702672","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-06-25 14:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-06-25/577385/4/check/tempest-dsvm-intel-nfv-xenial/e8ce2c1 : SUCCESS in 30m 21s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-25/577385/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/932009a : FAILURE in 1h 30m 52s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-25/577385/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/69ac2df : SUCCESS in 57m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a62637042e1cbedebd01fa09fa3156c19cc3040","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-26 01:07:06.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/577385/4/ : FAILURE in 10h 05m 08s","accounts_in_message":[],"_revision_number":4},{"id":"74538c2d99ecdc4375e3022f3a03da121ba4e2d9","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-06-26 08:02:56.000000000","message":"Patch Set 4: Code-Review-1\n\nAlso with the note, your should update some tests that refer you changes.","accounts_in_message":[],"_revision_number":4},{"id":"a8baf449e21790a82d52f6a053fc3044b87c6043","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-26 12:00:22.000000000","message":"Patch Set 4:\n\n\u003e \u003e Don\u0027t you need to add a reno note when you change the default\n \u003e \u003e behavior of an option?\n \u003e \n \u003e Oh, good point. There should probably be a small reno for this, yes\n\nSure, I\u0027ll write up a quick one.","accounts_in_message":[],"_revision_number":4},{"id":"6dbf65a87c1e99ab3f8e45edb3cf6c6aad5f17c4","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-06-26 15:57:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3fc1e8a4d1a01eba08e572e6bd2f96a53de7bb4d","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-06-26 16:22:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fd3d6fbfaf70f4f9bcf4302fc8f76f44904bffd2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-06-26 17:45:52.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/5/check/tempest-dsvm-full-xenial/f058b32/ : SUCCESS in 1h 09m 13s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/85/577385/5/check/grenade-dsvm-xenial/11dd024/ : SUCCESS in 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c87eebd66c3b4e246f951554890621754fde27f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-26 18:12:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/577385/5/check/openstack-tox-pep8/3764151/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/85/577385/5/check/openstack-tox-py27/7871949/ : SUCCESS in 15m 22s\n- openstack-tox-py35 http://logs.openstack.org/85/577385/5/check/openstack-tox-py35/44fe5aa/ : SUCCESS in 14m 10s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/577385/5/check/build-openstack-sphinx-docs/c5da4a6/html/ : SUCCESS in 7m 32s\n- tempest-full http://logs.openstack.org/85/577385/5/check/tempest-full/6122e8d/ : SUCCESS in 1h 40m 05s\n- neutron-grenade http://logs.openstack.org/85/577385/5/check/neutron-grenade/11db95f/ : SUCCESS in 53m 15s\n- tempest-full-py3 http://logs.openstack.org/85/577385/5/check/tempest-full-py3/9913108/ : SUCCESS in 1h 24m 34s\n- build-openstack-releasenotes http://logs.openstack.org/85/577385/5/check/build-openstack-releasenotes/1b771b4/html/ : SUCCESS in 16m 18s\n- neutron-grenade-multinode http://logs.openstack.org/85/577385/5/check/neutron-grenade-multinode/cff0340/ : SUCCESS in 55m 43s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/85/577385/5/check/legacy-grenade-dsvm-neutron-multinode-live-migration/1f4778e/ : SUCCESS in 55m 56s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/85/577385/5/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/03656d9/ : SUCCESS in 1h 12m 01s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/85/577385/5/check/neutron-tempest-multinode-full/ccea721/ : SUCCESS in 1h 39m 34s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/85/577385/5/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1448145/ : SUCCESS in 45m 43s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/85/577385/5/check/nova-cells-v1/edc3825/ : SUCCESS in 55m 48s\n- nova-live-migration http://logs.openstack.org/85/577385/5/check/nova-live-migration/167979c/ : SUCCESS in 43m 20s\n- nova-multiattach http://logs.openstack.org/85/577385/5/check/nova-multiattach/df6aa61/ : SUCCESS in 57m 14s\n- nova-next http://logs.openstack.org/85/577385/5/check/nova-next/0dba158/ : SUCCESS in 1h 22m 24s\n- nova-tox-functional http://logs.openstack.org/85/577385/5/check/nova-tox-functional/b88e0ee/ : SUCCESS in 18m 05s\n- nova-tox-functional-py35 http://logs.openstack.org/85/577385/5/check/nova-tox-functional-py35/6013edd/ : SUCCESS in 18m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/577385/5/check/openstack-tox-lower-constraints/39938a6/ : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":5},{"id":"1b7394719ea5eac6a1a22907df039ec1a8c882a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-06-26 18:27:27.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/85/577385/5/check-vote/ext-nova-zuul/a616409 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":5},{"id":"de7f462d060daef0c72fa61f51746471347305f7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-06-26 18:38:54.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-20739 : FAILURE in 58m 43s","accounts_in_message":[],"_revision_number":5},{"id":"699cc9d5f68aee9694085ad59a268d9201def7bf","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-06-26 19:11:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/85/577385/5/check/dsvm-tempest-neutron-network/2b9e520 : SUCCESS in 2h 30m 08s","accounts_in_message":[],"_revision_number":5},{"id":"4c91590fa89ec3728539959b0bea12e7b9ae1b0d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-06-26 19:32:29.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/577385/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"3a4f1369d331d482ae29820bb65ee36246082253","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-06-26 19:34:37.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/85/577385/5/check/nova-out-of-tree-pvm/cbf7e01 : FAILURE in 1h 38m 08s\n- nova-in-tree-pvm http://184.172.12.213/85/577385/5/check/nova-in-tree-pvm/5c92659 : FAILURE in 1h 42m 03s","accounts_in_message":[],"_revision_number":5},{"id":"e332031bf6f34a272cf73723ac2b4db369d836f4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-06-26 19:48:34.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/577385/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7c80123 : SUCCESS in 1h 33m 14s","accounts_in_message":[],"_revision_number":5},{"id":"731b042ada06676d5b98a381eaa3c9121298e69c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-06-26 21:16:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-06-26/577385/5/check/tempest-dsvm-intel-nfv-xenial/8822693 : SUCCESS in 41m 21s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-26/577385/5/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/c7fa9d7 : FAILURE in 1h 33m 02s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-06-26/577385/5/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/53d3ca7 : SUCCESS in 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d656823bc8ca5841d698ac2bbc4f6dd54cd4a7db","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-06-27 04:26:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/85/577385/5/check-nova/Nova-ML2-Sriov/e0240d7 : SUCCESS in 54m 29s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/85/577385/5/check-nova/Nova-MACVTAP-ML2-Sriov/cf30cf8 : SUCCESS in 43m 38s (non-voting)\n- NVMe http://13.74.249.42/85/577385/5/check-nova/NVMe/4946fce : SUCCESS in 46m 26s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"c279490e4463eb85bcfd803162c463d56e5afc48","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2018-06-27 06:16:39.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e9f59767c40e5147a483f7ee0be1ea4176182789","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-06-27 06:42:24.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Hyper-V CI, add a review comment with check hyper-v\n\n- nova-dsvm-full-tempest http://cloudbase-ci.com//nova/577385/5/ : FAILURE in 4h 53m 53s","accounts_in_message":[],"_revision_number":5},{"id":"4eff0c03700409c042f0e90d82d4ceb01cac2a75","author":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"date":"2018-06-27 06:43:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9dc7e8e1825ba32dfb702a8e0cc037d76ce734c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2018-06-27 15:00:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"55346753e305c4ba030087680b07303f82e46999","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-28 22:10:39.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nSure enough from the man page:\n\n   Usage\n       If you are unsure about whether you should use /dev/random or /dev/urandom, then probably you want to use the latter.  As a general rule, /dev/urandom should  be\n       used for everything except long-lived GPG/SSL/SSH keys.","accounts_in_message":[],"_revision_number":5},{"id":"6c408779d733c56d52534a6497cd533be9c7fc24","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-28 22:14:09.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nCan you also put up a docs patch to glance for updating the image properties for this?","accounts_in_message":[],"_revision_number":5},{"id":"cbf86f39acd31ac38d208277b9051d2b3443649b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-28 22:16:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f030b2715c7ffc0cb9ed9d68070d4566f17ae06d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-06-28 22:16:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ddb2b6eb4053da914c0e1ff99eccbcbce55d0a81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 03:52:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a914bc7594b91d3b8ddb6176668744a9c14732d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 07:18:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/85/577385/5/gate/openstack-tox-pep8/2b6eb61/ : SUCCESS in 10m 55s\n- openstack-tox-py27 http://logs.openstack.org/85/577385/5/gate/openstack-tox-py27/ac7d35e/ : SUCCESS in 16m 24s\n- openstack-tox-py35 http://logs.openstack.org/85/577385/5/gate/openstack-tox-py35/0db06cb/ : SUCCESS in 11m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/577385/5/gate/build-openstack-sphinx-docs/8b61a56/html/ : SUCCESS in 6m 35s\n- tempest-full http://logs.openstack.org/85/577385/5/gate/tempest-full/958099d/ : SUCCESS in 1h 35m 38s\n- neutron-grenade http://logs.openstack.org/85/577385/5/gate/neutron-grenade/62f0fcd/ : SUCCESS in 55m 38s\n- tempest-full-py3 http://logs.openstack.org/85/577385/5/gate/tempest-full-py3/3c35e0b/ : SUCCESS in 1h 29m 12s\n- build-openstack-releasenotes http://logs.openstack.org/85/577385/5/gate/build-openstack-releasenotes/50c85b7/html/ : SUCCESS in 22m 17s\n- nova-cells-v1 http://logs.openstack.org/85/577385/5/gate/nova-cells-v1/5ca67f7/ : SUCCESS in 47m 42s\n- nova-live-migration http://logs.openstack.org/85/577385/5/gate/nova-live-migration/9bbe1e4/ : SUCCESS in 46m 08s\n- nova-multiattach http://logs.openstack.org/85/577385/5/gate/nova-multiattach/9807df6/ : SUCCESS in 1h 00m 23s\n- nova-next http://logs.openstack.org/85/577385/5/gate/nova-next/eafc302/ : SUCCESS in 1h 24m 40s\n- nova-tox-functional http://logs.openstack.org/85/577385/5/gate/nova-tox-functional/a34941e/ : SUCCESS in 17m 38s\n- nova-tox-functional-py35 http://logs.openstack.org/85/577385/5/gate/nova-tox-functional-py35/1c51fed/ : SUCCESS in 16m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/577385/5/gate/openstack-tox-lower-constraints/fde446d/ : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":5},{"id":"3587cc4b46ec43b8f551e2defa0f7cae197e902a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-29 07:18:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"b162e52cfa0b8febefca28797e1ea44c42ef1716","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-07-02 12:22:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9dd16a53aa9e31d122d709337d79fcf406416a45","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-07-02 12:43:31.000000000","message":"Patch Set 5:\n\nUpdated the Glance documentation:\n\n    https://review.openstack.org/#/c/577385/\n    useful-image-properties.rst: Update default RNG source","accounts_in_message":[],"_revision_number":5},{"id":"2d8af7480d878f34655939e10fdf0663c3ea6117","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-07-02 13:38:07.000000000","message":"Patch Set 5:\n\n\u003e Updated the Glance documentation:\n\nCorrect link: https://review.openstack.org/#/c/579537/","accounts_in_message":[],"_revision_number":5},{"id":"1b54d88c1d0e1a9ef4ea26bb3b64fc7a5d5f2bbb","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2018-07-03 07:54:39.000000000","message":"Patch Set 3:\n\n\u003e \u003e Updated the Glance documentation:\n \u003e \n \u003e Correct link: https://review.openstack.org/#/c/579537/\n\nOops, thanks.","accounts_in_message":[],"_revision_number":3},{"id":"4403de04805e6276292228d9309396182d328b3e","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-05 13:31:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 5080076f6056e98dd492894c7f976b1b58aaf44a","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"814bfd937238cbd211ea30805c36ae682cfd7b48","revisions":{"98e05985114853a926faaeec66d3c1890e8a97e1":{"kind":"REWORK","_number":1,"created":"2018-06-22 10:17:24.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/85/577385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/577385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/577385/1"}}},"commit":{"parents":[{"commit":"2fb5be45e697bdceea9573e60b8b7417e40a5d4b","subject":"Merge \"objects: Remove NUMATopologyLimits.obj_from_db_obj\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fb5be45e697bdceea9573e60b8b7417e40a5d4b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:11:56.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:16:55.000000000","tz":120},"subject":"conf: libvirt: Update the outdated `rng_dev_path` docs","message":"conf: libvirt: Update the outdated `rng_dev_path` docs\n\nSince libvirt 1.3.4 any RNG device is accepted.  However, the\nrecommended source of entropy is `/dev/urandom` (since it doesn\u0027t have\nthe same limitations of `dev/random`).\n\nUpdate the documention to reflect that.\n\nChange-Id: Ia39402a045ffb1943463b5741655d84071613e8c\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\nReported-by: Daniel P. Berrangé \u003cberrange@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98e05985114853a926faaeec66d3c1890e8a97e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98e05985114853a926faaeec66d3c1890e8a97e1"}]},"branch":"refs/heads/master"},"90e4dc823193f939894874befef6617147117f14":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-06-22 10:18:45.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/85/577385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/577385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/577385/2"}}},"commit":{"parents":[{"commit":"2fb5be45e697bdceea9573e60b8b7417e40a5d4b","subject":"Merge \"objects: Remove NUMATopologyLimits.obj_from_db_obj\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fb5be45e697bdceea9573e60b8b7417e40a5d4b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:11:56.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:18:36.000000000","tz":120},"subject":"conf: libvirt: Update the outdated `rng_dev_path` docs","message":"conf: libvirt: Update the outdated `rng_dev_path` docs\n\nSince libvirt 1.3.4 any RNG device is accepted.  However, the\nrecommended source of entropy is `/dev/urandom` (as it doesn\u0027t have the\nsame limitations of `dev/random`; and doesn\u0027t block as easily).\n\nUpdate the documention to reflect that.\n\nChange-Id: Ia39402a045ffb1943463b5741655d84071613e8c\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\nReported-by: Daniel P. Berrangé \u003cberrange@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90e4dc823193f939894874befef6617147117f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90e4dc823193f939894874befef6617147117f14"}]},"branch":"refs/heads/master"},"32986d1cad981e0322b898426973f8bb0b552984":{"kind":"REWORK","_number":3,"created":"2018-06-22 15:51:26.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/85/577385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/577385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/577385/3"}}},"commit":{"parents":[{"commit":"2fb5be45e697bdceea9573e60b8b7417e40a5d4b","subject":"Merge \"objects: Remove NUMATopologyLimits.obj_from_db_obj\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fb5be45e697bdceea9573e60b8b7417e40a5d4b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:11:56.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 15:49:52.000000000","tz":120},"subject":"conf: libvirt: Update the outdated docs for `rng_dev_path`","message":"conf: libvirt: Update the outdated docs for `rng_dev_path`\n\nSince libvirt 1.3.4, any RNG device path (that returns random numbers\nwhen read!) is accepted.  However, the recommended source of entropy is\n`/dev/urandom` (it is non-blocking; and  doesn\u0027t have the same\nlimitations of `dev/random`, which is a legacy interface).\n\nUpdate the documention to reflect that.\n\nChange-Id: Ia39402a045ffb1943463b5741655d84071613e8c\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\nReported-by: Daniel P. Berrangé \u003cberrange@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32986d1cad981e0322b898426973f8bb0b552984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32986d1cad981e0322b898426973f8bb0b552984"}]},"branch":"refs/heads/master"},"05041a7434478c38fe7eb5a18f49450c0daf3fb9":{"kind":"REWORK","_number":4,"created":"2018-06-25 12:09:27.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/85/577385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/577385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/577385/4"}}},"commit":{"parents":[{"commit":"2fb5be45e697bdceea9573e60b8b7417e40a5d4b","subject":"Merge \"objects: Remove NUMATopologyLimits.obj_from_db_obj\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fb5be45e697bdceea9573e60b8b7417e40a5d4b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:11:56.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-25 12:06:38.000000000","tz":120},"subject":"conf: libvirt: Make `/dev/urandom` the default for \u0027rng_dev_path\u0027","message":"conf: libvirt: Make `/dev/urandom` the default for \u0027rng_dev_path\u0027\n\nSince libvirt 1.3.4, any RNG (Random Number Generator) device path (that\nreturns random numbers when read!) is accepted.  However, the\nrecommended source of entropy is `/dev/urandom` (it is non-blocking; and\ndoesn\u0027t have the same limitations of `dev/random`, which is a legacy\ninterface).\n\nTherefore, make `/dev/urandom` the default RNG for \u0027rng_dev_path\u0027 config\nattribute.  And update the documention to reflect that.\n\nChange-Id: Ia39402a045ffb1943463b5741655d84071613e8c\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\nReported-by: Daniel P. Berrangé \u003cberrange@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05041a7434478c38fe7eb5a18f49450c0daf3fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05041a7434478c38fe7eb5a18f49450c0daf3fb9"}]},"branch":"refs/heads/master"},"814bfd937238cbd211ea30805c36ae682cfd7b48":{"kind":"REWORK","_number":5,"created":"2018-06-26 15:57:19.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/85/577385/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/85/577385/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/85/577385/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/85/577385/5"}}},"commit":{"parents":[{"commit":"2fb5be45e697bdceea9573e60b8b7417e40a5d4b","subject":"Merge \"objects: Remove NUMATopologyLimits.obj_from_db_obj\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2fb5be45e697bdceea9573e60b8b7417e40a5d4b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-22 10:11:56.000000000","tz":120},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2018-06-26 15:55:11.000000000","tz":120},"subject":"conf: libvirt: Make `/dev/urandom` the default for \u0027rng_dev_path\u0027","message":"conf: libvirt: Make `/dev/urandom` the default for \u0027rng_dev_path\u0027\n\nSince libvirt 1.3.4, any RNG (Random Number Generator) device path (that\nreturns random numbers when read!) is accepted.  However, the\nrecommended source of entropy is `/dev/urandom` (it is non-blocking; and\ndoesn\u0027t have the same limitations of `dev/random`, which is a legacy\ninterface).\n\nTherefore, make `/dev/urandom` the default RNG for \u0027rng_dev_path\u0027 config\nattribute; adjust the relevant tests.  Also update the documention to\nreflect this change.\n\nChange-Id: Ia39402a045ffb1943463b5741655d84071613e8c\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\nReported-by: Daniel P. Berrangé \u003cberrange@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/814bfd937238cbd211ea30805c36ae682cfd7b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/814bfd937238cbd211ea30805c36ae682cfd7b48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
