)]}'
{"id":"openstack%2Fnova~590041","triplet_id":"openstack%2Fnova~master~If9ddb8a6d2f03392f3cc11136c4a0b026212b95b","project":"openstack/nova","branch":"master","topic":"one-resource-per-query","hashtags":[],"change_id":"If9ddb8a6d2f03392f3cc11136c4a0b026212b95b","subject":"[placement] split gigantor SQL query, add logging","status":"MERGED","created":"2018-08-08 21:13:30.000000000","updated":"2018-08-28 08:23:26.000000000","submitted":"2018-08-27 18:16:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"590041-1535393788681-e8143908","meta_rev_id":"e9f8ad5140e99dafabfc7b66a22981c9e617da9f","_number":590041,"virtual_id_number":590041,"owner":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-08-27 15:31:06.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"date":"2018-08-27 16:32:54.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 15:40:15.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 17:06:06.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 15:54:27.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 15:07:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 14:45:07.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2018-08-27 18:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2018-08-27 15:53:36.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2018-08-27 14:33:01.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2018-08-27 18:20:05.000000000","post_submit":true,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2018-08-27 16:42:22.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2018-08-27 16:22:26.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"recommended":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-08-27 14:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2018-08-27 13:51:52.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2018-08-27 13:51:52.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-09 20:15:54.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-08-09 20:36:27.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2018-08-12 13:04:49.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2018-08-14 16:13:30.000000000","updated_by":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"reviewer":{"_account_id":21813,"name":"Andrey Volkov","email":"m@amadev.ru","username":"avolkov"},"state":"REVIEWER"},{"updated":"2018-08-14 16:14:40.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-08-14 22:15:05.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2018-08-15 15:06:24.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-08-15 22:31:28.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2018-08-16 05:43:12.000000000","updated_by":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"reviewer":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"state":"REVIEWER"},{"updated":"2018-08-24 15:05:58.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-08-24 15:44:21.000000000","updated_by":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"reviewer":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"state":"REVIEWER"},{"updated":"2018-08-24 21:25:23.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2018-08-24 23:22:48.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2018-08-27 00:34:50.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2018-08-27 06:07:16.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-08-27 13:51:43.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2018-08-27 13:51:52.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2018-08-27 14:01:36.000000000","updated_by":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"reviewer":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"state":"REVIEWER"},{"updated":"2018-08-27 14:33:01.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-08-27 14:45:07.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 15:07:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 15:31:06.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 15:40:15.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 15:53:36.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 15:54:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 16:22:26.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 16:32:54.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 16:42:22.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 17:06:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 18:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-27 18:20:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c54e1ccaf159d109406dd344110ff5b95115bf34","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-08 21:13:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9ecd949faaf5deb5117667afc720334d3e7601c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-08 21:47:44.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/590041/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b6e191144da5c75a752813443ffd45c36ddf36cc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-08 22:08:17.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/1/check-vote/ext-nova-zuul/2e9a01b : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":1},{"id":"87758ea80322d4ffdabdd39b173aec6a7e3d3030","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-08 22:19:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/1/check/tempest-dsvm-full-xenial/6a45ef0/ : SUCCESS in 1h 05m 43s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/1/check/grenade-dsvm-xenial/496d4ba/ : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b811a4a10a6e3d6e099e115a5d0720381f67227","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-08 22:22:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/41/590041/1/check/nova-out-of-tree-pvm/2ca4445 : SUCCESS in 1h 08m 18s\n- nova-in-tree-pvm http://184.172.12.213/41/590041/1/check/nova-in-tree-pvm/48683ee : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7a88aec5daeb72e9bea97953a501c959e0512085","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-08 22:45:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-08/590041/1/check/tempest-dsvm-intel-nfv-xenial/b2b1088 : SUCCESS in 33m 30s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-08/590041/1/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/acbb87f : FAILURE in 1h 31m 17s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-08/590041/1/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/019faaa : SUCCESS in 1h 02m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03577ef4dd53bdc2925c527ea627c9c04b705d03","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-08 23:04:43.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d76d28b : FAILURE in 1h 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f290149e359c62a558a5a017df35882ced9efccf","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-08 23:19:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/1 : SUCCESS in 2h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5eda40a296f3d3ffd4f8194c002aaac83c4b110c","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-08 23:40:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/1/check/dsvm-tempest-neutron-network/569ddf8 : SUCCESS in 2h 26m 08s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/1/check/test-vgpu/fba1a99 : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"506b7b5b9c1bd56c61417307880c07e67aae1533","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-08 23:44:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/1/check/check-dsvm-tempest-vz7-exe-minimal/ff00808 : SUCCESS in 59m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"951411f8934e334d1ce7572159c8b81b41887ee4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-09 01:23:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/1/check-nova/Nova-ML2-Sriov/711f1fb : SUCCESS in 55m 57s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/1/check-nova/Nova-MACVTAP-ML2-Sriov/87ba0e7 : SUCCESS in 54m 46s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"4a6025be0b9e3303409f10d6f7a3abe685a9a264","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 01:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/1/check/openstack-tox-pep8/684df5b/ : SUCCESS in 9m 29s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/1/check/openstack-tox-py27/5421d8e/ : SUCCESS in 13m 00s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/1/check/openstack-tox-py35/11c88d6/ : SUCCESS in 13m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/1/check/build-openstack-sphinx-docs/f5e01a7/html/ : SUCCESS in 7m 32s\n- tempest-full http://logs.openstack.org/41/590041/1/check/tempest-full/92d93e2/ : SUCCESS in 1h 31m 02s\n- neutron-grenade http://logs.openstack.org/41/590041/1/check/neutron-grenade/68da8e3/ : SUCCESS in 46m 43s\n- tempest-full-py3 http://logs.openstack.org/41/590041/1/check/tempest-full-py3/db08dec/ : SUCCESS in 1h 22m 23s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/1/check/neutron-grenade-multinode/f982ea5/ : SUCCESS in 56m 34s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/4c8e7c8/ : SUCCESS in 1h 07m 49s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/def670c/ : SUCCESS in 1h 06m 33s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/1/check/neutron-tempest-multinode-full/61cd2a6/ : SUCCESS in 2h 01m 55s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2fd634a/ : SUCCESS in 43m 06s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/1/check/nova-cells-v1/7578d81/ : SUCCESS in 1h 03m 17s\n- nova-live-migration http://logs.openstack.org/41/590041/1/check/nova-live-migration/aaeefc4/ : SUCCESS in 53m 56s\n- nova-multiattach http://logs.openstack.org/41/590041/1/check/nova-multiattach/5c580b1/ : SUCCESS in 54m 31s\n- nova-next http://logs.openstack.org/41/590041/1/check/nova-next/d487e1e/ : SUCCESS in 1h 31m 56s\n- tempest-slow http://logs.openstack.org/41/590041/1/check/tempest-slow/fdb625b/ : SUCCESS in 1h 34m 30s\n- nova-tox-functional http://logs.openstack.org/41/590041/1/check/nova-tox-functional/702c08c/ : SUCCESS in 16m 05s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/1/check/nova-tox-functional-py35/139340e/ : SUCCESS in 16m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/1/check/openstack-tox-lower-constraints/3994f88/ : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"eba37c370ae4ca0adf03991c8a6fdfd28e5583e4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-09 04:43:28.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22544 : SUCCESS in 1h 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4e05645a0da1d85bcba97c47845aa775aeba4cee","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-08-09 07:36:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis increases the numbers of SQL queries, but since\n1. the increase delta doesn\u0027t depend on the cloud scale,\n2. this provides us both a big improvement on code readability and a crisp way to debug\nThis looks great to me. Thank you so much.\n\nJust one request inline.","accounts_in_message":[],"_revision_number":1},{"id":"d46a33e65649a17c389fab68c0dbe0ede8c1db03","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-08-09 09:25:17.000000000","message":"Patch Set 1:\n\ni\u0027ll run my little perf explore against this later in the day when I\u0027m back near my puter","accounts_in_message":[],"_revision_number":1},{"id":"797cf1904c060ba2471caab9a55f7dba5915fe2e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-09 15:22:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7352eeab9935ebfd36de588fef891944a401eb3b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-09 15:49:48.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/590041/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ac2d6eb4458ae78f93a61694493914b060ed026d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-09 15:56:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9a3e6eea47314d690d0632c8e2093a3d646a2595","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-09 16:44:32.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/3/check-vote/ext-nova-zuul/b7acf54 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":3},{"id":"de20f65765bb46876f6192780932f71f75b65cc2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-09 17:00:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/3/check/tempest-dsvm-full-xenial/09be97e/ : SUCCESS in 1h 02m 49s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/3/check/grenade-dsvm-xenial/989fabb/ : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eef5c4e9adffffc565706b90b29b8cc856eef241","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 17:23:26.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/73138fc : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":3},{"id":"0e8c313504eccb0a18ed8c1070774288cca908f1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-09 17:26:24.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22581 : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":3},{"id":"9a0d06b3a5fd0f8433743b741cd5ca4e5f3f36c8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-09 17:32:08.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/590041/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ff15cd334111755a56369d5e94f8bfdeb26a7cd8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 18:03:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/3/check/check-dsvm-tempest-vz7-exe-minimal/f4f4bda : SUCCESS in 1h 01m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":3},{"id":"aeacd39c6d6a122a8c0de7f908b0169eed571325","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-09 18:21:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/3 : SUCCESS in 2h 23m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ed6a360dd0475c6cf4e9113b150bdb4c31ec6865","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 18:33:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-09/590041/3/check/tempest-dsvm-intel-nfv-xenial/894073d : SUCCESS in 32m 48s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-09/590041/3/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/57df652 : FAILURE in 1h 31m 16s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-09/590041/3/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/6e07794 : SUCCESS in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c157afe54f4a1887c7007d95f87a42a854f453e","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-09 18:51:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/3/check/dsvm-tempest-neutron-network/89026d8 : SUCCESS in 2h 40m 35s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/3/check/test-vgpu/1690c02 : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"761077aab3fbe4f4c55ec7456b59e2b45c255d71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 18:52:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/3/check/openstack-tox-pep8/d7d2e58/ : SUCCESS in 11m 21s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/3/check/openstack-tox-py27/d89185a/ : SUCCESS in 11m 19s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/3/check/openstack-tox-py35/ef427cb/ : SUCCESS in 12m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/3/check/build-openstack-sphinx-docs/34872bd/html/ : SUCCESS in 7m 14s\n- tempest-full http://logs.openstack.org/41/590041/3/check/tempest-full/6bf1bdb/ : SUCCESS in 1h 32m 04s\n- neutron-grenade http://logs.openstack.org/41/590041/3/check/neutron-grenade/2213e0d/ : SUCCESS in 46m 20s\n- tempest-full-py3 http://logs.openstack.org/41/590041/3/check/tempest-full-py3/546c2ae/ : SUCCESS in 1h 25m 34s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/3/check/neutron-grenade-multinode/9813a06/ : SUCCESS in 59m 28s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/da10f48/ : SUCCESS in 58m 31s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/3/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/53415cd/ : SUCCESS in 1h 48m 56s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/3/check/neutron-tempest-multinode-full/0b3df3b/ : SUCCESS in 2h 01m 05s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/3/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/e2dafc3/ : SUCCESS in 52m 41s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/3/check/nova-cells-v1/d1757e6/ : SUCCESS in 46m 31s\n- nova-live-migration http://logs.openstack.org/41/590041/3/check/nova-live-migration/7096261/ : FAILURE in 42m 40s\n- nova-multiattach http://logs.openstack.org/41/590041/3/check/nova-multiattach/388680c/ : SUCCESS in 1h 06m 44s\n- nova-next http://logs.openstack.org/41/590041/3/check/nova-next/09d86b6/ : FAILURE in 2h 20m 12s\n- tempest-slow http://logs.openstack.org/41/590041/3/check/tempest-slow/3d7ea91/ : SUCCESS in 1h 40m 52s\n- nova-tox-functional http://logs.openstack.org/41/590041/3/check/nova-tox-functional/a80ce3f/ : SUCCESS in 15m 45s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/3/check/nova-tox-functional-py35/e875913/ : SUCCESS in 24m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/3/check/openstack-tox-lower-constraints/8091b52/ : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"2df7ebc6351854efcde37e02e277f72ac139f9a2","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-08-09 20:15:54.000000000","message":"Patch Set 3:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- nova-live-migration: https://bugs.launchpad.net/bugs/1761798\n- nova-next: https://bugs.launchpad.net/bugs/1737039\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"51654b199d40c975f9300960050289c4612a86dd","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-09 20:17:22.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"03bacd9b3eed21eaa67911018d08e098c93bf3e4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-09 20:47:31.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/590041/4 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"590a3acfb80111def0326a2a0575ac30541ffdbc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-09 21:16:36.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/4/check-vote/ext-nova-zuul/9b6d003 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":4},{"id":"b5ce3dc068481bd6c73d6ed76293b7ea090b9737","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 21:23:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/4/check/check-dsvm-tempest-vz7-exe-minimal/36539ee : SUCCESS in 1h 03m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"d685edc2f9531b2935f5d49ae7cfcdbd70b34421","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-09 21:27:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/4/check/tempest-dsvm-full-xenial/dadc0bb/ : SUCCESS in 1h 09m 57s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/4/check/grenade-dsvm-xenial/bbb449b/ : SUCCESS in 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d3093d1a5e4d2f5198a54b0f220d3fcb036f7c8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 21:39:58.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f4dcd84 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":4},{"id":"c144f9c345bcf38afb08fbb170e6147d418d764d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-09 21:51:34.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22593 : SUCCESS in 1h 33m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d992aba964ca9f4a65e235be09114f60bf8b0fca","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-09 21:55:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-09/590041/4/check/tempest-dsvm-intel-nfv-xenial/32313ea : SUCCESS in 1h 00m 53s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-09/590041/4/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/22f05fc : FAILURE in 1h 31m 31s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-09/590041/4/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/614b4b2 : SUCCESS in 1h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"26601a5bb864d0f1c6ad142e105251b36d43aa52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 22:16:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/4/check/openstack-tox-pep8/2cf2230/ : SUCCESS in 10m 59s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/4/check/openstack-tox-py27/e0986f0/ : SUCCESS in 13m 40s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/4/check/openstack-tox-py35/632e839/ : SUCCESS in 14m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/4/check/build-openstack-sphinx-docs/59dd01f/html/ : SUCCESS in 8m 46s\n- tempest-full http://logs.openstack.org/41/590041/4/check/tempest-full/16ea96a/ : SUCCESS in 1h 39m 50s\n- neutron-grenade http://logs.openstack.org/41/590041/4/check/neutron-grenade/16253e1/ : SUCCESS in 1h 01m 30s\n- tempest-full-py3 http://logs.openstack.org/41/590041/4/check/tempest-full-py3/faec7b9/ : SUCCESS in 1h 18m 59s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/4/check/neutron-grenade-multinode/d2f12eb/ : SUCCESS in 1h 05m 57s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/c8dc894/ : SUCCESS in 1h 04m 43s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/4/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/3f2ea94/ : SUCCESS in 1h 07m 02s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/4/check/neutron-tempest-multinode-full/fd6c9d2/ : SUCCESS in 1h 49m 32s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/4/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2607be6/ : SUCCESS in 45m 17s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/4/check/nova-cells-v1/2a35fba/ : SUCCESS in 55m 59s\n- nova-live-migration http://logs.openstack.org/41/590041/4/check/nova-live-migration/62d3892/ : SUCCESS in 40m 44s\n- nova-multiattach http://logs.openstack.org/41/590041/4/check/nova-multiattach/3ce0642/ : SUCCESS in 59m 29s\n- nova-next http://logs.openstack.org/41/590041/4/check/nova-next/096f454/ : SUCCESS in 1h 38m 29s\n- tempest-slow http://logs.openstack.org/41/590041/4/check/tempest-slow/af2d646/ : SUCCESS in 1h 41m 32s\n- nova-tox-functional http://logs.openstack.org/41/590041/4/check/nova-tox-functional/332a34b/ : SUCCESS in 19m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/4/check/nova-tox-functional-py35/220f30b/ : SUCCESS in 15m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/4/check/openstack-tox-lower-constraints/a0c2ce2/ : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c0157a4acb9607710c282c258ed6af230e86e4b3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-09 22:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/4 : SUCCESS in 2h 10m 36s","accounts_in_message":[],"_revision_number":4},{"id":"c1dc740632afe6c7be8eaaaf0438739b0d55e1d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-09 22:44:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/4/check/dsvm-tempest-neutron-network/50c27f2 : SUCCESS in 2h 24m 48s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/4/check/test-vgpu/7499092 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68a1177744faca777eea3ce7dfe0382bc8271c12","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-10 02:06:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/4/check-nova/Nova-ML2-Sriov/fd18acf : SUCCESS in 1h 09m 31s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/4/check-nova/Nova-MACVTAP-ML2-Sriov/2d4a750 : SUCCESS in 49m 56s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":4},{"id":"9974e2766d7d526ac656879252446efae8eb648f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-08-10 08:53:50.000000000","message":"Patch Set 4: Code-Review+1\n\nwfm.\n\nOne thing we might want to consider is if all these log messages should be at level debug. Given what people have said on the related email thread, people (sadly) want this all the time but there are mixed messages about whether logging should be set to debugging always. I don\u0027t know the history of all that, so I\u0027m not sure I understand the nuances, but thought it worth mentioning.","accounts_in_message":[],"_revision_number":4},{"id":"09d09e8c28495b68339691f3356f307e52395514","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-10 14:44:36.000000000","message":"Patch Set 4:\n\nChris tested this in his perf harness yesterday [1] and the slow-down with this split-out was negligible (\u003c0.5%, which may even be within the bounds of normal variance).\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-placement/%23openstack-placement.2018-08-09.log.html#t2018-08-09T16:10:43","accounts_in_message":[],"_revision_number":4},{"id":"3ccd5b775638664238c3637484f983533be269ed","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-08-10 14:52:53.000000000","message":"Patch Set 4:\n\n\u003e wfm.\n \u003e \n \u003e One thing we might want to consider is if all these log messages\n \u003e should be at level debug. Given what people have said on the\n \u003e related email thread, people (sadly) want this all the time but\n \u003e there are mixed messages about whether logging should be set to\n \u003e debugging always. I don\u0027t know the history of all that, so I\u0027m not\n \u003e sure I understand the nuances, but thought it worth mentioning.\n\nI think they should be at level debug based on my experience working on a large-scale deployment at my last company. We weren\u0027t able to run the scheduler at log level DEBUG all the time because it slowed the scheduler down significantly. So logging all of that at INFO, I would be wary of. That said, I\u0027m not sure what the differences would be in this particular case DEBUG vs INFO. Maybe it\u0027s not nearly as significant as it was in the scheduler filters case.","accounts_in_message":[],"_revision_number":4},{"id":"72bdc4eabaa1b78deba5f679c9a4fdfbb826c3e7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-10 16:07:31.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"50a29497fc925992dab88f06f87ea7d4a770c5a4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 16:24:15.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/5/check/check-dsvm-tempest-vz7-exe-minimal/4584fe5 : FAILURE in 11m 12s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"a8138e86c340b31cb31448311d47a7a3b236c763","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 16:59:56.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/590041/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"ce18395136fb3503595145ce72fe81661d22cdf6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-10 17:11:57.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/5/check-vote/ext-nova-zuul/c9fd018 : FAILURE in 54m 09s","accounts_in_message":[],"_revision_number":5},{"id":"8331c839c2f089fc7f6397b9c60436f5eba9820a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-10 17:12:07.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/5/check/tempest-dsvm-full-xenial/cab0505/ : SUCCESS in 1h 03m 44s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/5/check/grenade-dsvm-xenial/1a4209d/ : SUCCESS in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58674d262175cca2540775ae0d014abeb3678ab0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-10 17:21:19.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22629 : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":5},{"id":"523e2956142b72be24681650c00fcd19367d21c7","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-08-10 17:35:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2d6a99b7ce15cd045a44009f8c54e80063a2edba","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 18:11:29.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/6/check/check-dsvm-tempest-vz7-exe-minimal/33dc75c : FAILURE in 15m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"753d4954d8288f1173cf9bf48c5e2f9a1955e0b3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 18:31:45.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/590041/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"fb785804e6bbae43d963daf84400d631ff6c195d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-10 18:32:46.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/6/check-vote/ext-nova-zuul/dc6ed33 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":6},{"id":"00b131dbf6303c4b08b311175e5f4c34c359c26d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-10 18:33:20.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/6/check/tempest-dsvm-full-xenial/269ebfd/ : SUCCESS in 55m 03s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/6/check/grenade-dsvm-xenial/41391ee/ : SUCCESS in 56m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"570d7b41914ebd8afa645b5a33e826690f8d36cf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-10 18:45:27.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22633 : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":6},{"id":"c2bd90ace17f8e08f0d7eb864cb6844e27fe74a2","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 18:57:33.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/8911d26 : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":6},{"id":"9ad26db696ba5a325701dcd4fcb90dd10a46a9e3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 19:13:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-10/590041/6/check/tempest-dsvm-intel-nfv-xenial/2a79fc1 : SUCCESS in 31m 53s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-10/590041/6/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9abf810 : FAILURE in 1h 31m 42s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-10/590041/6/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/5e3d9a2 : SUCCESS in 1h 00m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b353db801a37817080dcb4039a6f46b7c18451ba","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-10 19:39:38.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/6/check-nova/Nova-ML2-Sriov/21ff09b : SUCCESS in 1h 15m 22s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/6/check-nova/Nova-MACVTAP-ML2-Sriov/974991d : SUCCESS in 39m 35s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":6},{"id":"d6807b5d072b123abd4a0eae445d7339526ba7dd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-10 19:42:33.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/6 : SUCCESS in 2h 05m 14s","accounts_in_message":[],"_revision_number":6},{"id":"3605f0c0d548abd964c2fbd86d959b01d7f6f5ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-10 19:57:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/6/check/openstack-tox-pep8/a03ca29/ : SUCCESS in 9m 49s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/6/check/openstack-tox-py27/f0e83b5/ : SUCCESS in 13m 06s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/6/check/openstack-tox-py35/2179122/ : SUCCESS in 15m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/6/check/build-openstack-sphinx-docs/52b26f5/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/41/590041/6/check/tempest-full/eefbc8a/ : SUCCESS in 1h 44m 01s\n- neutron-grenade http://logs.openstack.org/41/590041/6/check/neutron-grenade/6e9007d/ : SUCCESS in 50m 31s\n- tempest-full-py3 http://logs.openstack.org/41/590041/6/check/tempest-full-py3/b97ffa9/ : FAILURE in 1h 21m 24s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/6/check/neutron-grenade-multinode/74c30f4/ : SUCCESS in 1h 01m 51s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/6/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b0d9be6/ : SUCCESS in 1h 10m 16s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/6/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/0b89e0e/ : SUCCESS in 1h 16m 20s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/6/check/neutron-tempest-multinode-full/4a8479e/ : SUCCESS in 1h 47m 58s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/6/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/29852f8/ : SUCCESS in 44m 10s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/6/check/nova-cells-v1/a71a31c/ : SUCCESS in 55m 41s\n- nova-live-migration http://logs.openstack.org/41/590041/6/check/nova-live-migration/41f361b/ : SUCCESS in 46m 05s\n- nova-multiattach http://logs.openstack.org/41/590041/6/check/nova-multiattach/93121a8/ : SUCCESS in 1h 02m 17s\n- nova-next http://logs.openstack.org/41/590041/6/check/nova-next/f2a15a8/ : SUCCESS in 1h 31m 57s\n- tempest-slow http://logs.openstack.org/41/590041/6/check/tempest-slow/db725d7/ : SUCCESS in 1h 30m 19s\n- nova-tox-functional http://logs.openstack.org/41/590041/6/check/nova-tox-functional/80970be/ : SUCCESS in 19m 47s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/6/check/nova-tox-functional-py35/07528c7/ : SUCCESS in 19m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/6/check/openstack-tox-lower-constraints/5cd5847/ : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":6},{"id":"29433e6cc26df8aab2ada93e5d47ab07697a1dbc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-10 20:06:38.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/6/check/dsvm-tempest-neutron-network/884332c : FAILURE in 2h 22m 36s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/6/check/test-vgpu/4949e04 : SUCCESS in 1m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"06bf28db0c39f2f6b1058b5e7aa828ce95b6d0a9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 20:11:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eded1d99d5b96e450877cc31cfd59878e8db6707","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-10 20:12:48.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5262acaea9b09c9c0ccf1069fdea9ecff46ba112","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 20:13:29.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"0a071ccd7ae0067eb6aeda5c0f143d40d57e80eb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-10 20:16:37.000000000","message":"Patch Set 8: Code-Review-1\n\nSee PS6 comments. https://review.openstack.org/#/c/590041/6/nova/api/openstack/placement/objects/resource_provider.py@2937 is the only remaining blocker (the messages are showing the wrong number as is).","accounts_in_message":[],"_revision_number":8},{"id":"d115b1f8757493277c5e0437fc3160972b9310c3","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 20:19:57.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"c4e70fb39e451d8e9a972dd9f3507958f8eaeb81","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 20:39:37.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/9/check/check-dsvm-tempest-vz7-exe-minimal/4fb9cdb : FAILURE in 15m 18s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"714d3317137bff7d3c14f7b96e621a9cb29ab285","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 20:55:53.000000000","message":"Patch Set 7:\n\n* pci-test http://52.27.155.124/pci/590041/7 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"c7de30f0a077781977c50f40e2b515303e54f9a7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-10 21:18:52.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/9/check-vote/ext-nova-zuul/22025ad : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":9},{"id":"f984d78d8153b55f6554206063755979e0218692","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 21:29:59.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"876b09d6dff94fb8fdd3c5ca875b8d279a5b3aa6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-10 21:37:43.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/9/check/tempest-dsvm-full-xenial/ef4e729/ : SUCCESS in 1h 16m 27s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/9/check/grenade-dsvm-xenial/0ab1f23/ : SUCCESS in 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4dd303e7374a8df1398057d46665ea6834a23c09","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 21:43:48.000000000","message":"Patch Set 9:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/9/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7a23736 : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":9},{"id":"44b1dc94fe2103bada16ff9e8cb33f84fe6e8622","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 21:47:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"428f0dd867da18ba0c4add9e3d1a36cd17bf0fc5","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 21:51:19.000000000","message":"Patch Set 8:\n\n* pci-test http://52.27.155.124/pci/590041/8 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"c2bff876a818a39777fb6e7043f41aded7709864","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-10 22:02:02.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4d3eac26f55efabdff94093d3e4350c3d1ecb92f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 22:10:54.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/10/check/check-dsvm-tempest-vz7-exe-minimal/402f124 : FAILURE in 14m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"805a188364580de21ec5617e238255358c0d0e69","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 22:30:33.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"006c216c1f62e469fb797496d4c2cf66ae4b604c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-10 22:31:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"932d597ffb94347bb6074fe05e7c622d027b4510","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-10 22:56:42.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/11/check/check-dsvm-tempest-vz7-exe-minimal/b073ccd : FAILURE in 19m 44s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"2485aaa1fea95c5beb7cc0403abef1cadb9d8d85","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 23:07:18.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/590041/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"7d956d73577963356a852eee891e6fa2f69ee774","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-10 23:26:02.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/11/check-vote/ext-nova-zuul/e49652c : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":11},{"id":"0cd5bfa5386cdbb0fa492b748e1a7ab60fc21060","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-10 23:26:11.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22652 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":11},{"id":"010d8927d5a405afdcea9d0bf773f9d4b0f23ef9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-10 23:36:47.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/11/check/tempest-dsvm-full-xenial/9b826a9/ : SUCCESS in 1h 05m 13s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/11/check/grenade-dsvm-xenial/0001169/ : SUCCESS in 55m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"88e40afa327fd5cb57bbd58c1e40341f64ffa64a","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-10 23:38:24.000000000","message":"Patch Set 10:\n\n* pci-test http://52.27.155.124/pci/590041/10 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"9367560270a617cb9f3fbf22d4ffbd5d90345fbf","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-11 00:03:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-10/590041/11/check/tempest-dsvm-intel-nfv-xenial/4d7eac8 : SUCCESS in 30m 48s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-10/590041/11/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/334627d : FAILURE in 1h 31m 49s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-10/590041/11/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/fa3061c : SUCCESS in 1h 00m 45s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ae7fdd300ed2a664fa248f24c0a23a692693dd5f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-11 00:03:59.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/a196470 : SUCCESS in 1h 32m 00s","accounts_in_message":[],"_revision_number":11},{"id":"71ecc627c8b4e37d771df96ffde729ee943f0867","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-11 00:13:36.000000000","message":"Patch Set 11:\n\n* pci-test http://52.27.155.124/pci/590041/11 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"def270995c041f0ac35862cf8464b5fac9823f91","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-11 00:36:18.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/11 : SUCCESS in 2h 04m 30s","accounts_in_message":[],"_revision_number":11},{"id":"2870840539fc263cbcff3318149918ed9c03b8e3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-11 00:40:52.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/11/check/dsvm-tempest-neutron-network/26d1445 : SUCCESS in 2h 08m 38s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/11/check/test-vgpu/7ff7e65 : SUCCESS in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"142503baac5dcb348ca8159a5f9dbcf78ea07d8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-11 01:08:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/11/check/openstack-tox-pep8/6ab749c/ : SUCCESS in 10m 05s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/11/check/openstack-tox-py27/d434cd8/ : SUCCESS in 12m 09s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/11/check/openstack-tox-py35/9a7ae16/ : SUCCESS in 14m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/11/check/build-openstack-sphinx-docs/22fe876/html/ : SUCCESS in 7m 27s\n- tempest-full http://logs.openstack.org/41/590041/11/check/tempest-full/dd7a82b/ : SUCCESS in 1h 31m 55s\n- neutron-grenade http://logs.openstack.org/41/590041/11/check/neutron-grenade/de2be59/ : SUCCESS in 56m 24s\n- tempest-full-py3 http://logs.openstack.org/41/590041/11/check/tempest-full-py3/af2347a/ : SUCCESS in 1h 27m 13s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/11/check/neutron-grenade-multinode/0b68699/ : SUCCESS in 1h 05m 22s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/11/check/legacy-grenade-dsvm-neutron-multinode-live-migration/e53174f/ : SUCCESS in 1h 06m 33s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/11/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/ac42012/ : SUCCESS in 1h 14m 54s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/11/check/neutron-tempest-multinode-full/643c1a3/ : SUCCESS in 1h 56m 04s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/11/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a8f7723/ : SUCCESS in 49m 23s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/11/check/nova-cells-v1/a2dd1fb/ : SUCCESS in 52m 30s\n- nova-live-migration http://logs.openstack.org/41/590041/11/check/nova-live-migration/f1f5250/ : SUCCESS in 47m 18s\n- nova-multiattach http://logs.openstack.org/41/590041/11/check/nova-multiattach/46d68a2/ : SUCCESS in 57m 00s\n- nova-next http://logs.openstack.org/41/590041/11/check/nova-next/5a041c1/ : SUCCESS in 1h 34m 50s\n- tempest-slow http://logs.openstack.org/41/590041/11/check/tempest-slow/0d9eede/ : SUCCESS in 1h 27m 03s\n- nova-tox-functional http://logs.openstack.org/41/590041/11/check/nova-tox-functional/bb1ac84/ : SUCCESS in 18m 03s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/11/check/nova-tox-functional-py35/075061b/ : SUCCESS in 16m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/11/check/openstack-tox-lower-constraints/c6257a7/ : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":11},{"id":"ddda35f06340c2fd212c2f62a332ef671c3010e0","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-11 01:59:25.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/11/check-nova/Nova-ML2-Sriov/863b93b : SUCCESS in 1h 05m 05s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/11/check-nova/Nova-MACVTAP-ML2-Sriov/d4893c4 : SUCCESS in 39m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"e761eec3e83cab1c66826af6ae94be40d4892597","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-08-11 07:44:58.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nForbidden trait filtering is negative(-) filter while the other filters are all positive(\u0026) filter. In principle, we\u0027d like to allocate positive filters first and then negative filters. Shuffling them sounds bugbear to me.","accounts_in_message":[],"_revision_number":11},{"id":"5a7c6aea5392a6b82332711c36bc0910474fa0df","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-11 13:27:24.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)\n\nIn terms of code readability, I think we hit the peak at PS7. My suggestions there led to more efficiency, but less readability. We could continue to do more of that and make the code as efficient as possible (short of adding agg/trait filtering to _get_providers_with_resource()), but we shouldn\u0027t.\n\nAt this point my beef is with the debug logging, which was one of the main purposes of this effort. Within the algorithm as currently written, a suggestion for the frequency/granularity (if not exact wording :) of the debug messages is noted at [1].\n\n[1] https://review.openstack.org/#/c/590041/10/nova/api/openstack/placement/objects/resource_provider.py@2885","accounts_in_message":[],"_revision_number":11},{"id":"702cea788cbb2f5f85b72537c084608f9bc8d2bc","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-12 12:53:05.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f81d4714a59faab0a08449f81c23ab68f3a3331e","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-12 13:02:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3c6a24e340a9410886544ec7db2c99ec4af56f0e","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2018-08-12 13:04:49.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"879e34dca63dfcc70f846b9abcf71169b03e13dd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-13 13:11:59.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"693b257bbf2e698f000262960895c42cc661683b","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-13 13:24:55.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"836d3e94c29e5b5c48f559d96fe42f4a192b793c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-13 13:25:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"569e6214c1aa18650d5ecc3bd562dd253643d823","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-13 16:07:57.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nTo reiterate, my downvote at this point is not for logic tweaks, but for the debug messages.","accounts_in_message":[],"_revision_number":11},{"id":"1672b48c9a533093fa03be03a7ab881df4f17fa2","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-13 16:15:55.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"37a9b8888039eda57be81080aa9fbc5332875682","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-13 19:36:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8624c647f0e4d3d62c76910ab7495589606c0ee0","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-13 19:47:34.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nI\u0027m actually quite confused what everyone is asking me to change about this at this point.","accounts_in_message":[],"_revision_number":11},{"id":"10595e15c2a3b1c815bea0103f30e9e23ab1cd35","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-13 21:13:00.000000000","message":"Patch Set 11:\n\n\u003e I\u0027m actually quite confused what everyone is asking me to change about this at this point.\n\nRealizing there was confusion, I threw a patch on top [1] to clarify. Note that it incorporates optimizations of code paths (but not code lines) mentioned here as optional fixups, but the main thrust (and the only thing I\u0027m downvoting this patch for) is the debug messages.\n\n[1] https://review.openstack.org/#/c/591510/","accounts_in_message":[],"_revision_number":11},{"id":"802ff3266a6d8ee7fab30c56a5b1cf3e9b718ece","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-14 15:37:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fb1c2c6ecd2a653f6742a4b1942063ba4eefe6c3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-14 15:38:56.000000000","message":"Patch Set 12: Code-Review+1\n\nPer Jay\u0027s request, squashed https://review.openstack.org/#/c/591510/ into this one.\n\n+1 approving the bits I didn\u0027t write :)","accounts_in_message":[],"_revision_number":12},{"id":"51f0a9192bf53745ab608d085ebfd8df06c68b31","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-14 15:39:13.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/590041/12 : FAILURE in 37s","accounts_in_message":[],"_revision_number":12},{"id":"62def19be778465344020c9fcc28c61b670ac13e","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-14 15:45:16.000000000","message":"Patch Set 12: Code-Review+1\n\nI like it. Not +2\u0027ing since I wrote a chunk of it.","accounts_in_message":[],"_revision_number":12},{"id":"d2d66c7d6ff5c8a90af41530f64a483c14685bc5","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-08-14 15:51:17.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nwfm","accounts_in_message":[],"_revision_number":12},{"id":"0d027169861c3e0f4b55df41f42e334a0f5679a2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-14 16:20:24.000000000","message":"Patch Set 12:\n\n* pci-test http://52.27.155.124/pci/590041/12 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"86b071aebe98da118c3ba9cfc14600fe391a6c39","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-14 16:23:17.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/12/check/check-dsvm-tempest-vz7-exe-minimal/6173836 : FAILURE in 12m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"77eeecba30e9267bd8cb3deeef9675aa8c7c340b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-14 16:50:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/41/590041/12/check/nova-out-of-tree-pvm/0ed43d6 : SUCCESS in 1h 08m 43s\n- nova-in-tree-pvm http://184.172.12.213/41/590041/12/check/nova-in-tree-pvm/68c9e9f : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":12},{"id":"01053e13bfb75595cbc04c9ff020f352cad26eb5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-14 16:50:49.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/12/check/tempest-dsvm-full-xenial/5906969/ : SUCCESS in 1h 12m 37s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/12/check/grenade-dsvm-xenial/c502913/ : SUCCESS in 59m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7280eecb585f535c650e2ebe155b2ac0fe001a69","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-14 16:55:58.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22732 : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":12},{"id":"f162d2ba26a9646de865d298a6e8fed19cd31db1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-14 17:23:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-14/590041/12/check/tempest-dsvm-intel-nfv-xenial/3f0ae86 : SUCCESS in 35m 41s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-14/590041/12/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9cc9318 : FAILURE in 1h 31m 37s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-14/590041/12/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/2190ead : SUCCESS in 1h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"315b24db1bb4733e4c326deb3c1b533c73395822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-14 17:42:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/12/check/openstack-tox-pep8/73aa6b8/ : SUCCESS in 9m 55s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/12/check/openstack-tox-py27/c4f5b8e/ : SUCCESS in 12m 24s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/12/check/openstack-tox-py35/77280bd/ : SUCCESS in 11m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/12/check/build-openstack-sphinx-docs/1c2032a/html/ : SUCCESS in 6m 57s\n- tempest-full http://logs.openstack.org/41/590041/12/check/tempest-full/6d78b36/ : SUCCESS in 1h 54m 32s\n- neutron-grenade http://logs.openstack.org/41/590041/12/check/neutron-grenade/009b9d0/ : SUCCESS in 50m 59s\n- tempest-full-py3 http://logs.openstack.org/41/590041/12/check/tempest-full-py3/45d9ad1/ : SUCCESS in 1h 29m 46s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/12/check/neutron-grenade-multinode/5531756/ : SUCCESS in 1h 04m 42s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/12/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b6f8980/ : SUCCESS in 1h 07m 03s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/12/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/2792536/ : SUCCESS in 1h 09m 18s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/12/check/neutron-tempest-multinode-full/62bb85e/ : SUCCESS in 1h 56m 09s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/12/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9632043/ : SUCCESS in 52m 36s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/12/check/nova-cells-v1/7cd645d/ : SUCCESS in 53m 13s\n- nova-live-migration http://logs.openstack.org/41/590041/12/check/nova-live-migration/185b6a4/ : SUCCESS in 48m 24s\n- nova-multiattach http://logs.openstack.org/41/590041/12/check/nova-multiattach/650cb6d/ : SUCCESS in 1h 02m 57s\n- nova-next http://logs.openstack.org/41/590041/12/check/nova-next/f1e07e4/ : SUCCESS in 1h 34m 51s\n- tempest-slow http://logs.openstack.org/41/590041/12/check/tempest-slow/5e908ba/ : SUCCESS in 1h 36m 58s\n- nova-tox-functional http://logs.openstack.org/41/590041/12/check/nova-tox-functional/23444ae/ : SUCCESS in 17m 22s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/12/check/nova-tox-functional-py35/ef5ae1f/ : SUCCESS in 18m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/12/check/openstack-tox-lower-constraints/4dc68e4/ : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":12},{"id":"cd35b3c603e08fdf015fb65c83dee4e41aa01c3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-08-14 17:42:36.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/590041/12/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":12},{"id":"01474698fa5cc352725cc8310c2a07b21e0629dc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-14 18:17:50.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/12/check/dsvm-tempest-neutron-network/b3e9c6b : SUCCESS in 2h 20m 14s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/12/check/test-vgpu/ec77a86 : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4daceaea16d97888605718da5c61eb6c7dd09a8b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-08-14 18:27:38.000000000","message":"Patch Set 12:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-590041-12 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":12},{"id":"ecf1e65dc067a0f806a7ab2ad8d428004a12f8cb","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-14 18:31:15.000000000","message":"Patch Set 12:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ceae51d : SUCCESS in 1h 49m 18s","accounts_in_message":[],"_revision_number":12},{"id":"bb83d6774160dc6d575862d9a36d0c70167a4be8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-08-15 03:30:46.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nLGTM. Thank you!!","accounts_in_message":[],"_revision_number":12},{"id":"a87bf9adf4c80752b5a4b9c94a8b4e803a85c639","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-15 07:03:19.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/12/check-vote/ext-nova-zuul/b55fe3d : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":12},{"id":"3744767bdeabfc1d889256e6edf92c3dc4dd7833","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-15 08:58:22.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"7b2fa5fd30004bf4b20b25cbae574136ddf0f681","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-15 12:53:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"559dd3bc645e84eda54c5064802002e27d3f76a0","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-15 12:56:47.000000000","message":"Patch Set 12: Code-Review+1\n\nDown to +1, since I saw Eric\u0027s email, let us see if there is any feedback first.","accounts_in_message":[],"_revision_number":12},{"id":"a542dc478df8b0f928d127cbe261507c546beb8c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-15 13:14:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"18012b2927ad9c63e4e2f42ec81e6b84c4f28e54","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-15 13:17:18.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"1565116dceae5d0cfda35d93d163272f953efcd2","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-15 13:25:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"21a7663d3eda3039996601a196f2a235e2508f78","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-15 13:28:12.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4f56febb5425423523ab0639c0ef471184a0fa1a","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-15 13:29:13.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"461ef2d118a862da1cced0a65132f07cb1646a13","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-15 13:30:42.000000000","message":"Patch Set 15: Code-Review+1\n\nApprove-worthy; not +2ing because co-authored.","accounts_in_message":[],"_revision_number":15},{"id":"1b037e8dcbfeb0f6e289c0d18afb991f9940c953","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-15 13:50:47.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/15/check/check-dsvm-tempest-vz7-exe-minimal/834ca18 : FAILURE in 9m 41s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"0f9e72aebf01e75b914edb263d9bafd746663084","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-15 13:54:06.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/590041/13 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"6bd27b3deaa06d702a28ee8990f9f04ae7fb19d4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-15 13:55:53.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/590041/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"e56d26e1b5293cb6d62f363ce0e324d68be44520","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-15 14:07:07.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/590041/15 : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"ab51ccdf57dbf05f73d6dd53544b9aafc6c7f1f3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-15 14:20:03.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/15/check-vote/ext-nova-zuul/e6e1772 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":15},{"id":"d3ea35290a602b6f997035730e63177fda2458b9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-15 14:39:38.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/41/590041/15/check/nova-out-of-tree-pvm/6358f01 : SUCCESS in 1h 09m 42s\n- nova-in-tree-pvm http://184.172.12.213/41/590041/15/check/nova-in-tree-pvm/301914c : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":15},{"id":"06322c36ae683091b0f7327655560e0fec7c9768","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-15 14:40:36.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-22773 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":15},{"id":"f4c5d7d2e72c2c463422016a3c8fe5226d78eb99","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-15 15:02:18.000000000","message":"Patch Set 15:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/eaad46d : SUCCESS in 1h 30m 58s","accounts_in_message":[],"_revision_number":15},{"id":"a55da138312d9bbef90c4956c166efe5704adb5d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-15 15:02:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-15/590041/15/check/tempest-dsvm-intel-nfv-xenial/919127c : SUCCESS in 31m 34s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-15/590041/15/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/9d113f2 : FAILURE in 1h 32m 11s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-15/590041/15/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/9266f90 : SUCCESS in 1h 03m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"81adb4eacae0e5227f87d58de7d69064e49de3f9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-15 15:05:44.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/15/check/tempest-dsvm-full-xenial/9479e5b/ : SUCCESS in 1h 17m 37s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/15/check/grenade-dsvm-xenial/8d440d0/ : SUCCESS in 55m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"aea24ee06549fc6d14ea89e5730b50295c7dc981","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-08-15 15:06:24.000000000","message":"Patch Set 15: Code-Review+1\n\nlet\u0027s make this a thing","accounts_in_message":[],"_revision_number":15},{"id":"b5c4f4db81c16ab474258970f3cac3724ed03ebf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-15 15:35:22.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/15/check/openstack-tox-pep8/63f3fba/ : SUCCESS in 8m 47s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/15/check/openstack-tox-py27/a3d3b6b/ : SUCCESS in 12m 43s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/15/check/openstack-tox-py35/1224dd6/ : SUCCESS in 14m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/15/check/build-openstack-sphinx-docs/b46060c/html/ : SUCCESS in 7m 38s\n- tempest-full http://logs.openstack.org/41/590041/15/check/tempest-full/31d873e/ : SUCCESS in 2h 03m 58s\n- neutron-grenade http://logs.openstack.org/41/590041/15/check/neutron-grenade/a83019a/ : SUCCESS in 54m 42s\n- tempest-full-py3 http://logs.openstack.org/41/590041/15/check/tempest-full-py3/43b52b0/ : SUCCESS in 1h 20m 36s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/15/check/neutron-grenade-multinode/7d5e9d7/ : SUCCESS in 1h 01m 58s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/15/check/legacy-grenade-dsvm-neutron-multinode-live-migration/5c52b6a/ : SUCCESS in 1h 00m 31s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/15/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/ae7664a/ : SUCCESS in 1h 20m 59s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/15/check/neutron-tempest-multinode-full/0fb010e/ : SUCCESS in 1h 44m 34s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/15/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2011246/ : SUCCESS in 44m 48s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/15/check/nova-cells-v1/b401a96/ : SUCCESS in 1h 13m 16s\n- nova-live-migration http://logs.openstack.org/41/590041/15/check/nova-live-migration/7196d92/ : SUCCESS in 43m 34s\n- nova-multiattach http://logs.openstack.org/41/590041/15/check/nova-multiattach/61ca122/ : SUCCESS in 51m 50s\n- nova-next http://logs.openstack.org/41/590041/15/check/nova-next/d117f0a/ : SUCCESS in 1h 41m 51s\n- tempest-slow http://logs.openstack.org/41/590041/15/check/tempest-slow/e0c9eb8/ : SUCCESS in 1h 34m 49s\n- nova-tox-functional http://logs.openstack.org/41/590041/15/check/nova-tox-functional/382b846/ : SUCCESS in 16m 26s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/15/check/nova-tox-functional-py35/eccee76/ : SUCCESS in 14m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/15/check/openstack-tox-lower-constraints/e236f54/ : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":15},{"id":"9b889fbf09bb2f4dded84572daf0961fa711e045","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-08-15 15:35:44.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/590041/15/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"7206d1fb3ee6e5250840d7da3d555611c1ff4f43","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-15 15:47:24.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/15 : SUCCESS in 2h 17m 24s","accounts_in_message":[],"_revision_number":15},{"id":"76f5d49b016a90a9eeb74818ca331f3d8bd41607","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-08-15 16:26:00.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-590041-15 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":15},{"id":"61311e0a3dbd0e33fa98c7ee56ce66a6ea5d6403","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-08-15 16:51:39.000000000","message":"Patch Set 15:\n\nDon\u0027t know the code well enough to to have an informed opinion, just wanted to say that I think the general idea is good.","accounts_in_message":[],"_revision_number":15},{"id":"376e471e22cbcd2effb0b28a82c0b4a318ee5fc2","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-15 17:32:58.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/15/check/dsvm-tempest-neutron-network/aae9850 : FAILURE in 4h 01m 54s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/15/check/test-vgpu/a9f647e : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f6c94faeb662a93f84afff3b07e33a9dd1f23314","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2018-08-15 22:31:28.000000000","message":"Patch Set 15: Code-Review+1\n\nNot a placement expert obviously, but this looks great to me.  Thanks for the quick response!","accounts_in_message":[],"_revision_number":15},{"id":"d6154cae483a587e5bc20d302dc35cf89ddf65c5","author":{"_account_id":19944,"name":"Chen Hanxiao","email":"chen_han_xiao@126.com","username":"chenhanxiao"},"date":"2018-08-16 05:43:12.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"d863ec56c1b4a827eae7fc18812aaa6122ecc0f8","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-08-16 13:00:45.000000000","message":"Patch Set 15: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":15},{"id":"5ef1ca2970c7eaeb809478782db8325e1d82b5cb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-20 00:54:30.000000000","message":"Patch Set 15: Code-Review+2\n\nstill LGTM","accounts_in_message":[],"_revision_number":15},{"id":"3ccc76b7f9bbeb9fb93cb397ae3813e34158477b","author":{"_account_id":27738,"name":"mashuting","email":"shutingm@vmware.com","username":"mashuting"},"date":"2018-08-20 02:34:32.000000000","message":"Patch Set 15:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":15},{"id":"04840295b2d33cc0d0be37bfd82255d8fd8cec32","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-20 03:25:02.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/15/check-vote/ext-nova-zuul/31de67e : FAILURE in 49m 51s","accounts_in_message":[],"_revision_number":15},{"id":"4e68ed882db4a21cff1496fe4c4cbe77d1b83876","author":{"_account_id":27738,"name":"mashuting","email":"shutingm@vmware.com","username":"mashuting"},"date":"2018-08-20 05:15:21.000000000","message":"Patch Set 15:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":15},{"id":"464da7746ea265c75af41ee859659a874e972535","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-08-23 21:47:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cdcf26ab8d6656beadba6546b39da0efb13acdd0","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-08-23 22:57:41.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"40d3c193f46566bede706ac30d1e9b5489e37698","author":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"date":"2018-08-24 02:32:16.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"55335ce2924ea2ced66d73b79ab811ff023f55e1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-08-24 15:05:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"15d70ccdd8c57578d53db2720264b9ced486bfae","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-08-24 15:21:03.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"aca6f484be6e2cf774c28a1cb552faf83ecf4e56","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2018-08-24 15:35:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cb4ad5896e330c39c14e10bfc4c08ae8f7c2366f","author":{"_account_id":8768,"name":"Chris Friesen","email":"chris.friesen@windriver.com","username":"cbf123"},"date":"2018-08-24 15:44:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8a6c5f02f14d45b37d6131cda28504892c14160b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-24 19:14:37.000000000","message":"Patch Set 15: -Code-Review\n\n(1 comment)\n\nLooks like we need a rebase here anyway.","accounts_in_message":[],"_revision_number":15},{"id":"53d3df2220cf84603170c4bd6e150d12b3e291e3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-24 19:23:51.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"bd21c40c8407f24d8a97cb7c94030e5f53137faa","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-24 19:25:58.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"66128e4fe3b7424b5aec3711f25c99480d95b352","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-24 19:50:33.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/pci/590041/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"089338aad53dc0ebaeddee21dcde957a001efc4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-24 20:33:10.000000000","message":"Patch Set 16:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/16/check-vote/ext-nova-zuul/55b42c4 : FAILURE in 45m 08s","accounts_in_message":[],"_revision_number":16},{"id":"06588b1915c5332569b801f967f4622bf7982199","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-24 20:34:49.000000000","message":"Patch Set 16:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/16/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/70b7dca : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":16},{"id":"19d37cb8281a632fbf1f592abaca5b67062b8de3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-24 20:42:28.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/16/check/tempest-dsvm-full-xenial/9857855/ : SUCCESS in 1h 18m 09s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/16/check/grenade-dsvm-xenial/6c99583/ : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1aca8a88573ec8d061be216e5ce5413518525eaf","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-24 20:44:27.000000000","message":"Patch Set 16:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/41/590041/16/check/nova-out-of-tree-pvm/d267d86 : SUCCESS in 1h 20m 09s\n- nova-in-tree-pvm http://184.172.12.213/41/590041/16/check/nova-in-tree-pvm/f1c496e : FAILURE in 54m 02s","accounts_in_message":[],"_revision_number":16},{"id":"2529d037f6676893aa51f9c1c9bbce749a1e7a20","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-24 20:57:25.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-24/590041/16/check/tempest-dsvm-intel-nfv-xenial/30d6c2a : SUCCESS in 31m 10s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-24/590041/16/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/a66b794 : FAILURE in 1h 32m 48s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-24/590041/16/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/c887f3f : SUCCESS in 1h 00m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a940bf4b8ee850f03bdf9726e7ed49e1f9d1b920","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-24 21:03:08.000000000","message":"Patch Set 16:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/16/check/dsvm-tempest-neutron-network/31f7106 : FAILURE in 1h 38m 40s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/16/check/test-vgpu/e0e116f : SUCCESS in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"79c5b1583fca729b6c6e308405754c31a2a64fec","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-24 21:25:23.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-23155 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":16},{"id":"9ffb633d2294817142e7fce9a144b5f3342cff68","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-24 21:25:40.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/590041/16 : SUCCESS in 2h 01m 02s","accounts_in_message":[],"_revision_number":16},{"id":"8734ff88a4ffef332e98331c59aec68b08c9e48f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-24 21:36:26.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/16/check/openstack-tox-pep8/00feafb/ : SUCCESS in 9m 10s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/16/check/openstack-tox-py27/5918d2b/ : SUCCESS in 11m 55s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/16/check/openstack-tox-py35/296c250/ : SUCCESS in 11m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/16/check/build-openstack-sphinx-docs/393b6eb/html/ : SUCCESS in 6m 55s\n- tempest-full http://logs.openstack.org/41/590041/16/check/tempest-full/5363d6d/ : SUCCESS in 1h 39m 36s\n- neutron-grenade http://logs.openstack.org/41/590041/16/check/neutron-grenade/83ae848/ : SUCCESS in 49m 41s\n- tempest-full-py3 http://logs.openstack.org/41/590041/16/check/tempest-full-py3/0519ee3/ : SUCCESS in 1h 17m 18s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/16/check/neutron-grenade-multinode/ecb7426/ : SUCCESS in 1h 11m 04s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/16/check/legacy-grenade-dsvm-neutron-multinode-live-migration/403b110/ : SUCCESS in 55m 19s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/16/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/67e54a0/ : SUCCESS in 1h 23m 31s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/16/check/neutron-tempest-multinode-full/755a541/ : SUCCESS in 1h 55m 14s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/16/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7180ca2/ : SUCCESS in 42m 21s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/16/check/nova-cells-v1/750e562/ : SUCCESS in 57m 13s\n- nova-live-migration http://logs.openstack.org/41/590041/16/check/nova-live-migration/ccdbc5f/ : SUCCESS in 46m 57s\n- nova-multiattach http://logs.openstack.org/41/590041/16/check/nova-multiattach/cae4367/ : SUCCESS in 1h 01m 09s\n- nova-next http://logs.openstack.org/41/590041/16/check/nova-next/fe1d7f8/ : SUCCESS in 1h 40m 45s\n- tempest-slow http://logs.openstack.org/41/590041/16/check/tempest-slow/0437f45/ : SUCCESS in 1h 37m 09s\n- nova-tox-functional http://logs.openstack.org/41/590041/16/check/nova-tox-functional/d697683/ : SUCCESS in 15m 52s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/16/check/nova-tox-functional-py35/848aa4e/ : SUCCESS in 16m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/16/check/openstack-tox-lower-constraints/bff76b6/ : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":16},{"id":"cf7850d14dde01e41ffad1c41f7d350294c396a0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-08-24 22:44:26.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-590041-16 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":16},{"id":"7cbfed5685dcbc3d41132ed7437a9fedbab89340","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2018-08-24 23:22:48.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)\n\nLGTM, thanks y\u0027all","accounts_in_message":[],"_revision_number":16},{"id":"87b344a2ee87d8aed31298b09edf3bd72939150f","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-25 10:34:15.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/16/check-nova/Nova-ML2-Sriov/4d04714 : SUCCESS in 1h 06m 33s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/16/check-nova/Nova-MACVTAP-ML2-Sriov/16b268c : SUCCESS in 50m 55s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":16},{"id":"47621c7c985700dfddc109ac01a0a0ee55e6fe63","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-25 15:40:59.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4db6bfb1ab2bae84b034dbc9817d3eb4ea7ecf23","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2018-08-27 00:34:50.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"189e17d6688bd2471b6c56487acca3b99bf9df88","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-08-27 06:07:16.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"6c0fc52c4e3d3c99813e992183bf9723bb3b4cd9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-27 13:50:25.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"80ad4d747800f34457a04b02789ff99f8e3cd73b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2018-08-27 13:51:43.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"aeb15a3c844d575f1c59b58c20d1f520f64ea17c","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2018-08-27 13:51:52.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\nFast approve, since previous PS already +2ed by Melanie, the new PS just update the commit message.","accounts_in_message":[],"_revision_number":17},{"id":"35d0f051c76964b65cfb10a9e9848ac676a51069","author":{"_account_id":28239,"name":"zhaixiaojun","email":"zhaixiaojun@gohighsec.com","username":"zhaixiaojun"},"date":"2018-08-27 14:01:36.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"55194f6c04fe2b19163384f9a0b7955a6b470f3f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-08-27 14:33:01.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/590041/17 : FAILURE in 41m 41s","accounts_in_message":[],"_revision_number":17},{"id":"b404d33798acead8c3dcd29abf8ef87ab2f81249","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2018-08-27 14:45:07.000000000","message":"Patch Set 17:\n\n* pci-test http://52.27.155.124/pci/590041/17 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"cbe9e422a7d0bd7ccf68f3508078a47c01216a53","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-08-27 15:07:03.000000000","message":"Patch Set 17:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/17/check/tempest-dsvm-full-xenial/108ca00/ : SUCCESS in 1h 16m 03s\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/41/590041/17/check/grenade-dsvm-xenial/d29abaa/ : SUCCESS in 57m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e4aaefd395f39902b00df36d1755d3743fb42f5f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-08-27 15:31:06.000000000","message":"Patch Set 17:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/41/590041/17/check/nova-out-of-tree-pvm/12533bf : FAILURE in 1h 23m 02s\n- nova-in-tree-pvm http://184.172.12.213/41/590041/17/check/nova-in-tree-pvm/dc4beac : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":17},{"id":"f5617b32e1b670b0680a929174214ddb2e93e79b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-08-27 15:40:15.000000000","message":"Patch Set 17:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/590041/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c03e2e4 : SUCCESS in 1h 41m 11s","accounts_in_message":[],"_revision_number":17},{"id":"8e78e3bee8f637550084f7612caf3883819ab9a5","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-08-27 15:53:36.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/17/check/dsvm-tempest-neutron-network/afadfe7 : FAILURE in 2h 02m 38s\n- test-vgpu http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/41/590041/17/check/test-vgpu/082d3e3 : SUCCESS in 1m 06s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f89f55ba7f7453eea290ea6c10c8b11034812d26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 15:54:06.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/17/check/openstack-tox-pep8/5a0ca1c/ : SUCCESS in 10m 54s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/17/check/openstack-tox-py27/e9bc1d2/ : SUCCESS in 18m 02s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/17/check/openstack-tox-py35/49f2ebf/ : SUCCESS in 21m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/17/check/build-openstack-sphinx-docs/1955008/html/ : SUCCESS in 7m 29s\n- tempest-full http://logs.openstack.org/41/590041/17/check/tempest-full/b3f9ddd/ : SUCCESS in 1h 30m 44s\n- neutron-grenade http://logs.openstack.org/41/590041/17/check/neutron-grenade/7e8ae6d/ : SUCCESS in 58m 31s\n- tempest-full-py3 http://logs.openstack.org/41/590041/17/check/tempest-full-py3/7f3f4ba/ : SUCCESS in 1h 40m 52s\n- neutron-grenade-multinode http://logs.openstack.org/41/590041/17/check/neutron-grenade-multinode/a09a30f/ : SUCCESS in 1h 13m 31s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/41/590041/17/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b7a8dfc/ : SUCCESS in 56m 14s (non-voting)\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/41/590041/17/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/53f0446/ : SUCCESS in 1h 11m 11s (non-voting)\n- neutron-tempest-multinode-full http://logs.openstack.org/41/590041/17/check/neutron-tempest-multinode-full/653e5b6/ : SUCCESS in 1h 46m 29s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/41/590041/17/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0fef0c9/ : SUCCESS in 45m 30s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/41/590041/17/check/nova-cells-v1/c81de6f/ : SUCCESS in 1h 06m 56s\n- nova-live-migration http://logs.openstack.org/41/590041/17/check/nova-live-migration/dca4c0b/ : SUCCESS in 49m 01s\n- nova-multiattach http://logs.openstack.org/41/590041/17/check/nova-multiattach/5aa4938/ : SUCCESS in 55m 03s\n- nova-next http://logs.openstack.org/41/590041/17/check/nova-next/b65023d/ : SUCCESS in 1h 40m 01s\n- tempest-slow http://logs.openstack.org/41/590041/17/check/tempest-slow/dc146e1/ : SUCCESS in 1h 51m 58s\n- nova-tox-functional http://logs.openstack.org/41/590041/17/check/nova-tox-functional/d08e655/ : SUCCESS in 16m 24s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/17/check/nova-tox-functional-py35/805bd3d/ : SUCCESS in 15m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/17/check/openstack-tox-lower-constraints/c7a0acb/ : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":17},{"id":"6386f6e3fe5e63ce607e7ece37c18df169095ff6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2018-08-27 15:54:27.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/590041/17/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":17},{"id":"96308c1b8be3b7821f7d1fe1173145f60005506a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 15:55:56.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"3772c7cfc29cdce88bb39a523bdb49a627543963","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2018-08-27 16:22:26.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial http://52.27.155.124/portland/2018-08-27/590041/17/check/tempest-dsvm-intel-nfv-xenial/92ba12f : SUCCESS in 31m 23s (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-27/590041/17/check/tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial/e5925d2 : FAILURE in 1h 31m 59s (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial http://52.27.155.124/portland/2018-08-27/590041/17/check/tempest-dsvm-ovsdpdk-nfv-networking-xenial/cf345b3 : SUCCESS in 59m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"dc1f70e5b291bd61c8d135a7ec3cd50bc85f3e39","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2018-08-27 16:32:54.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/41/590041/17/check/check-dsvm-tempest-vz7-exe-minimal/4e481d3 : FAILURE in 20m 07s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"ecb862cea72e37c0633a0b17cdabbfb3699f6006","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2018-08-27 16:42:22.000000000","message":"Patch Set 17:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-41-590041-17 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":17},{"id":"1d7a9be9631496d22ccbeb82951e43b4eacfc2a9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2018-08-27 17:06:06.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/41/590041/17/check-nova/Nova-ML2-Sriov/a9e73b8 : FAILURE in 37m 58s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/41/590041/17/check-nova/Nova-MACVTAP-ML2-Sriov/880512c : FAILURE in 39m 59s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":17},{"id":"10b73039bb7da0041a3c1fa6ae9042011da7f903","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 18:16:28.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/41/590041/17/gate/openstack-tox-pep8/c36a7eb/ : SUCCESS in 10m 58s\n- openstack-tox-py27 http://logs.openstack.org/41/590041/17/gate/openstack-tox-py27/5463cbe/ : SUCCESS in 12m 58s\n- openstack-tox-py35 http://logs.openstack.org/41/590041/17/gate/openstack-tox-py35/505adcf/ : SUCCESS in 13m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/41/590041/17/gate/build-openstack-sphinx-docs/423f39b/html/ : SUCCESS in 8m 45s\n- tempest-full http://logs.openstack.org/41/590041/17/gate/tempest-full/818c989/ : SUCCESS in 1h 43m 57s\n- neutron-grenade http://logs.openstack.org/41/590041/17/gate/neutron-grenade/92315f1/ : SUCCESS in 54m 03s\n- tempest-full-py3 http://logs.openstack.org/41/590041/17/gate/tempest-full-py3/b39401b/ : SUCCESS in 1h 31m 23s\n- nova-cells-v1 http://logs.openstack.org/41/590041/17/gate/nova-cells-v1/5347086/ : SUCCESS in 1h 00m 34s\n- nova-live-migration http://logs.openstack.org/41/590041/17/gate/nova-live-migration/5d6a7e1/ : SUCCESS in 44m 10s\n- nova-multiattach http://logs.openstack.org/41/590041/17/gate/nova-multiattach/292b9cc/ : SUCCESS in 1h 23m 46s\n- nova-next http://logs.openstack.org/41/590041/17/gate/nova-next/e368430/ : SUCCESS in 1h 45m 21s\n- tempest-slow http://logs.openstack.org/41/590041/17/gate/tempest-slow/d9cd68c/ : SUCCESS in 1h 36m 25s\n- nova-tox-functional http://logs.openstack.org/41/590041/17/gate/nova-tox-functional/f2ad6f0/ : SUCCESS in 16m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/41/590041/17/gate/nova-tox-functional-py35/513d38f/ : SUCCESS in 15m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/41/590041/17/gate/openstack-tox-lower-constraints/4c7938b/ : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":17},{"id":"ac674313060c1db90e7c6b699d1b531b7e3456a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 18:16:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"80e5787a789a8a75512db797a4efbeaaa416bd5d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2018-08-27 18:20:05.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/41/590041/17/check-vote/ext-nova-zuul/a720de9 : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":17},{"id":"90f074f1a578fda3c6d47f8455ae06a20404f800","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2018-08-28 08:23:26.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-23214 : SUCCESS in 1h 55m 44s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"b5ab9f5acec172d16e46876f60ca338434483905","revisions":{"eb58da40ca97922285b5d7662fc1db57f9bda6d1":{"kind":"REWORK","_number":1,"created":"2018-08-08 21:13:30.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/1"}}},"commit":{"parents":[{"commit":"33415e53437bf0fde1b2bc1d87025c3ea93f2d74","subject":"Merge \"Refactor cell_type in compute/api.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33415e53437bf0fde1b2bc1d87025c3ea93f2d74"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested.\n\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb58da40ca97922285b5d7662fc1db57f9bda6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb58da40ca97922285b5d7662fc1db57f9bda6d1"}]},"branch":"refs/heads/master"},"5bd9fd20572aadccb8422a33e548a59873c9f46d":{"kind":"REWORK","_number":2,"created":"2018-08-09 15:22:58.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/2"}}},"commit":{"parents":[{"commit":"33415e53437bf0fde1b2bc1d87025c3ea93f2d74","subject":"Merge \"Refactor cell_type in compute/api.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33415e53437bf0fde1b2bc1d87025c3ea93f2d74"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-09 15:22:43.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bd9fd20572aadccb8422a33e548a59873c9f46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5bd9fd20572aadccb8422a33e548a59873c9f46d"}]},"branch":"refs/heads/master"},"6386619267d53319f0e632421be2f54ba3dfd81b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-08-09 15:56:44.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/3"}}},"commit":{"parents":[{"commit":"0346858c82073556d2937122190aa3be7a0a4741","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0346858c82073556d2937122190aa3be7a0a4741"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-09 15:55:39.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6386619267d53319f0e632421be2f54ba3dfd81b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6386619267d53319f0e632421be2f54ba3dfd81b"}]},"branch":"refs/heads/master"},"660723b60863eeff805b017750e943de240d6fb0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-08-09 20:17:22.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/4"}}},"commit":{"parents":[{"commit":"01e2ec61a0051b3ccc7b0fcd11dceea3c45f12a1","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01e2ec61a0051b3ccc7b0fcd11dceea3c45f12a1"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-09 18:20:16.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/660723b60863eeff805b017750e943de240d6fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/660723b60863eeff805b017750e943de240d6fb0"}]},"branch":"refs/heads/master"},"d861fc5724ae4103f165b321231e5b8241e32095":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-08-10 16:07:31.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/41/590041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/5"}}},"commit":{"parents":[{"commit":"37da760d008788c662a453a30d1ba78dff5e939d","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37da760d008788c662a453a30d1ba78dff5e939d"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-10 16:07:20.000000000","tz":-300},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d861fc5724ae4103f165b321231e5b8241e32095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d861fc5724ae4103f165b321231e5b8241e32095"}]},"branch":"refs/heads/master"},"2a10b87addae77e4a817a6a4bd712b017b25dd10":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2018-08-10 17:35:56.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/41/590041/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/6"}}},"commit":{"parents":[{"commit":"49a274d6e276eb27c04a6ffc5f94490fc6c2841b","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/49a274d6e276eb27c04a6ffc5f94490fc6c2841b"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2018-08-10 17:27:48.000000000","tz":60},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a10b87addae77e4a817a6a4bd712b017b25dd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a10b87addae77e4a817a6a4bd712b017b25dd10"}]},"branch":"refs/heads/master"},"660dc0176cd8ee0df29af5d2608a3be0862cdea8":{"kind":"REWORK","_number":7,"created":"2018-08-10 20:11:15.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/7"}}},"commit":{"parents":[{"commit":"49a274d6e276eb27c04a6ffc5f94490fc6c2841b","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/49a274d6e276eb27c04a6ffc5f94490fc6c2841b"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-10 20:10:42.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/660dc0176cd8ee0df29af5d2608a3be0862cdea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/660dc0176cd8ee0df29af5d2608a3be0862cdea8"}]},"branch":"refs/heads/master"},"92aab5b4d6e4b7d6cb3ed754068330a49bd4d0f3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2018-08-10 20:13:29.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/8"}}},"commit":{"parents":[{"commit":"b5883f3ea33d5a0aa08c30e86295b1b58a2eaf62","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5883f3ea33d5a0aa08c30e86295b1b58a2eaf62"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-10 20:12:25.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92aab5b4d6e4b7d6cb3ed754068330a49bd4d0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/92aab5b4d6e4b7d6cb3ed754068330a49bd4d0f3"}]},"branch":"refs/heads/master"},"7f90e2ddaac98a79d4173249cec31c8c7488257d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-08-10 20:19:57.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/9"}}},"commit":{"parents":[{"commit":"9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-10 20:19:42.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f90e2ddaac98a79d4173249cec31c8c7488257d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f90e2ddaac98a79d4173249cec31c8c7488257d"}]},"branch":"refs/heads/master"},"3ae57a5d8a561bbf5ed03a544863657cb2902ef6":{"kind":"REWORK","_number":10,"created":"2018-08-10 21:47:23.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/10"}}},"commit":{"parents":[{"commit":"9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-10 21:45:39.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ae57a5d8a561bbf5ed03a544863657cb2902ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ae57a5d8a561bbf5ed03a544863657cb2902ef6"}]},"branch":"refs/heads/master"},"f4ed285c029e8d7da4ebfdaa8344225e18fc6d87":{"kind":"REWORK","_number":11,"created":"2018-08-10 22:31:00.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/11"}}},"commit":{"parents":[{"commit":"9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa","subject":"placement: use simple code paths when possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ea340eb0d3bdb103bd64ca40b999bd2b10b80aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-10 22:30:45.000000000","tz":-240},"subject":"split gigantor SQL placement query into multiple","message":"split gigantor SQL placement query into multiple\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to the\n_get_providers_with_resource() calls, logging the number of matched\nproviders for each resource class requested and filter (on required\ntraits, forbidden traits and aggregate memebership).\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4ed285c029e8d7da4ebfdaa8344225e18fc6d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4ed285c029e8d7da4ebfdaa8344225e18fc6d87"}]},"branch":"refs/heads/master"},"4a93b316b9556df1a577d4da47765970daa95abd":{"kind":"REWORK","_number":12,"created":"2018-08-14 15:37:46.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/41/590041/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/12"}}},"commit":{"parents":[{"commit":"1316dd55e530446602ce174650949b0aa47893aa","subject":"Merge \"add zvm into support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1316dd55e530446602ce174650949b0aa47893aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-14 15:37:20.000000000","tz":-300},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n       5 after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n       2 after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ({\u0027HW_CPU_X86_AVX2\u0027: 65})\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ({u\u0027CUSTOM_TWO\u0027: 201, u\u0027CUSTOM_THREE\u0027: 202})\n found 3 providers with available 4 VCPU\n       1 after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a93b316b9556df1a577d4da47765970daa95abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a93b316b9556df1a577d4da47765970daa95abd"}]},"branch":"refs/heads/master"},"bc26e35b3b60be8bab95cc6ad81da66eabe5a49d":{"kind":"REWORK","_number":13,"created":"2018-08-15 13:14:03.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/13"}}},"commit":{"parents":[{"commit":"1316dd55e530446602ce174650949b0aa47893aa","subject":"Merge \"add zvm into support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1316dd55e530446602ce174650949b0aa47893aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-15 13:13:16.000000000","tz":-240},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n       5 after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n       2 after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ([\u0027HW_CPU_X86_AVX2\u0027])\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ([u\u0027CUSTOM_TWO\u0027, u\u0027CUSTOM_THREE\u0027])\n found 3 providers with available 4 VCPU\n       1 after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc26e35b3b60be8bab95cc6ad81da66eabe5a49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc26e35b3b60be8bab95cc6ad81da66eabe5a49d"}]},"branch":"refs/heads/master"},"06dce35b47b7c6b68ccbc107a3c73fbadbb84832":{"kind":"REWORK","_number":14,"created":"2018-08-15 13:25:52.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/14"}}},"commit":{"parents":[{"commit":"1316dd55e530446602ce174650949b0aa47893aa","subject":"Merge \"add zvm into support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1316dd55e530446602ce174650949b0aa47893aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-15 13:25:42.000000000","tz":-240},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n       5 after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n       2 after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ([\u0027HW_CPU_X86_AVX2\u0027])\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ([u\u0027CUSTOM_TWO\u0027, u\u0027CUSTOM_THREE\u0027])\n found 3 providers with available 4 VCPU\n       1 after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06dce35b47b7c6b68ccbc107a3c73fbadbb84832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06dce35b47b7c6b68ccbc107a3c73fbadbb84832"}]},"branch":"refs/heads/master"},"292b995d73bb84609a66f0491c563b7c04f7ae96":{"kind":"REWORK","_number":15,"created":"2018-08-15 13:29:13.000000000","uploader":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"ref":"refs/changes/41/590041/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/15"}}},"commit":{"parents":[{"commit":"1316dd55e530446602ce174650949b0aa47893aa","subject":"Merge \"add zvm into support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1316dd55e530446602ce174650949b0aa47893aa"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-15 13:29:05.000000000","tz":-240},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n       5 after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n       2 after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ([\u0027HW_CPU_X86_AVX2\u0027])\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ([u\u0027CUSTOM_TWO\u0027, u\u0027CUSTOM_THREE\u0027])\n found 3 providers with available 4 VCPU\n       1 after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/292b995d73bb84609a66f0491c563b7c04f7ae96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/292b995d73bb84609a66f0491c563b7c04f7ae96"}]},"branch":"refs/heads/master"},"b5afb6b171c08cb21856c614f664dc90c9a2dfbb":{"kind":"REWORK","_number":16,"created":"2018-08-24 19:23:51.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/41/590041/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/16"}}},"commit":{"parents":[{"commit":"8f0968d091c9debe1981ed5d72983627e27f3c55","subject":"Merge \"tests: Create functional libvirt test base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f0968d091c9debe1981ed5d72983627e27f3c55"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-24 19:23:41.000000000","tz":-300},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n       5 after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n       2 after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ([\u0027HW_CPU_X86_AVX2\u0027])\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ([u\u0027CUSTOM_TWO\u0027, u\u0027CUSTOM_THREE\u0027])\n found 3 providers with available 4 VCPU\n       1 after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5afb6b171c08cb21856c614f664dc90c9a2dfbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5afb6b171c08cb21856c614f664dc90c9a2dfbb"}]},"branch":"refs/heads/master"},"b5ab9f5acec172d16e46876f60ca338434483905":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2018-08-27 13:50:25.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/41/590041/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/41/590041/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/41/590041/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/41/590041/17"}}},"commit":{"parents":[{"commit":"8f0968d091c9debe1981ed5d72983627e27f3c55","subject":"Merge \"tests: Create functional libvirt test base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f0968d091c9debe1981ed5d72983627e27f3c55"}]}],"author":{"name":"Jay Pipes","email":"jaypipes@gmail.com","date":"2018-08-08 21:11:25.000000000","tz":-240},"committer":{"name":"Eric Fried","email":"efried@us.ibm.com","date":"2018-08-27 13:50:22.000000000","tz":0},"subject":"[placement] split gigantor SQL query, add logging","message":"[placement] split gigantor SQL query, add logging\n\nThis patch modifies the code paths for the non-granular request group\nallocation candidates processing. It removes the giant multi-join SQL\nquery and replaces it with multiple calls to\n_get_providers_with_resource(), logging the number of matched providers\nfor each resource class requested and filter (on required traits,\nforbidden traits and aggregate memebership).\n\nHere are some examples of the debug output:\n\n- A request for three resources with no aggregate or trait filters:\n\n found 7 providers with available 5 VCPU\n found 9 providers with available 1024 MEMORY_MB\n found 5 providers after filtering by previous result\n found 8 providers with available 1500 DISK_GB\n found 2 providers after filtering by previous result\n\n- The same request, but with a required trait that nobody has, shorts\n  out quickly:\n\n found 0 providers after applying required traits filter ([\u0027HW_CPU_X86_AVX2\u0027])\n\n- A request for one resource with aggregates and forbidden (but no\n  required) traits:\n\n found 2 providers after applying aggregates filter ([[\u00273ed8fb2f-4793-46ee-a55b-fdf42cb392ca\u0027]])\n found 1 providers after applying forbidden traits filter ([u\u0027CUSTOM_TWO\u0027, u\u0027CUSTOM_THREE\u0027])\n found 3 providers with available 4 VCPU\n found 1 providers after applying initial aggregate and trait filters\n\nCo-authored-by: Eric Fried \u003cefried@us.ibm.com\u003e\nCloses-Bug: #1786519\nChange-Id: If9ddb8a6d2f03392f3cc11136c4a0b026212b95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5ab9f5acec172d16e46876f60ca338434483905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5ab9f5acec172d16e46876f60ca338434483905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
