)]}'
{"id":"openstack%2Fnova~623510","triplet_id":"openstack%2Fnova~stable%2Fpike~Ie6da21e8680b2deb6b1da3add31cd725ba855c1c","project":"openstack/nova","branch":"stable/pike","topic":"bug/1806471","hashtags":[],"change_id":"Ie6da21e8680b2deb6b1da3add31cd725ba855c1c","subject":"Add secret\u003dtrue to fixed_key configuration parameter","status":"MERGED","created":"2018-12-07 14:52:27.000000000","updated":"2018-12-14 20:35:44.000000000","submitted":"2018-12-14 20:35:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"623510-1544819744959-87e4aefc","meta_rev_id":"776028c8bca9acafb6d8f775e6ddd9f91510c954","_number":623510,"virtual_id_number":623510,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-12-14 11:24:48.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2018-12-14 10:10:34.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2018-12-14 10:36:56.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},{"value":0,"date":"2018-12-14 10:33:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"value":2,"date":"2018-12-14 20:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2018-12-14 14:10:40.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2018-12-14 14:10:40.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-07 14:52:27.000000000","updated_by":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"reviewer":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"state":"REVIEWER"},{"updated":"2018-12-10 04:46:44.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2018-12-11 03:57:29.000000000","updated_by":{"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},"reviewer":{"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},"state":"REVIEWER"},{"updated":"2018-12-14 10:10:34.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2018-12-14 10:33:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-12-14 10:36:56.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-12-14 11:24:48.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-12-14 14:10:40.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-12-14 20:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"146068dfa885ac28318e4320c247b21bf44c3464","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2018-12-07 14:52:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e83b5cd9c0475fc5945f2b517351bcfae0692a85","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-12-07 15:37:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/623510/1 : FAILURE in 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9d64cc73b2df34465307db36b2443ac9606617f5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-07 15:56:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/10/623510/1/check/tempest-dsvm-full-xenial/0e6df18/ : SUCCESS in 1h 03m 32s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/10/623510/1/check/tempest-dsvm-full-xenial-py3/dbfa785/ : FAILURE in 57m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c442498c4e38aabe9f6dbb2e81da24d5a1dace80","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 16:25:33.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/623510/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e742285 : SUCCESS in 1h 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f927fbbebcc888f618bbb94db9167af67371f7be","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-12-07 18:18:58.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/10/623510/1/check/nova-out-of-tree-pvm/a7daa20 : SUCCESS in 1h 38m 11s\n- nova-in-tree-pvm http://184.172.12.213/10/623510/1/check/nova-in-tree-pvm/f73a59b : FAILURE in 3h 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1d28b352b612e1cb034cb8f29c7bbadb72d80da8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 21:54:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/623510/1/check/openstack-tox-pep8/a1f0a9a/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/10/623510/1/check/openstack-tox-py27/4f184f2/ : SUCCESS in 12m 30s\n- openstack-tox-py35 http://logs.openstack.org/10/623510/1/check/openstack-tox-py35/2b01049/ : SUCCESS in 11m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/10/623510/1/check/build-openstack-sphinx-docs/3830b54/html/ : SUCCESS in 5m 56s\n- tempest-full http://logs.openstack.org/10/623510/1/check/tempest-full/02c27d0/ : SUCCESS in 1h 29m 31s\n- neutron-grenade http://logs.openstack.org/10/623510/1/check/neutron-grenade/b55a759/ : SUCCESS in 56m 36s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/10/623510/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3ff3daf/ : SUCCESS in 41m 25s (non-voting)\n- nova-live-migration http://logs.openstack.org/10/623510/1/check/nova-live-migration/f6cc694/ : SUCCESS in 40m 22s\n- nova-tox-functional http://logs.openstack.org/10/623510/1/check/nova-tox-functional/7152046/ : SUCCESS in 18m 56s\n- nova-tox-functional-py35 http://logs.openstack.org/10/623510/1/check/nova-tox-functional-py35/ce7707d/ : SUCCESS in 17m 12s\n- neutron-grenade-multinode http://logs.openstack.org/10/623510/1/check/neutron-grenade-multinode/b4c1b6c/ : SUCCESS in 1h 00m 33s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/10/623510/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/470edc0/ : SUCCESS in 1h 00m 48s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/10/623510/1/check/devstack-plugin-ceph-tempest/4bbb56c/ : SUCCESS in 1h 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f1abe184cf33263e30ab89e074f2df0d6e6d270","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2018-12-10 04:46:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"922ea806320c3facb65e163879e860b1e19f303c","author":{"_account_id":28885,"name":"Ying Wang","email":"wangying@gohighsec.com","username":"yingwang"},"date":"2018-12-11 03:57:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f42f921408162009a7164de9cd5f11cd694da603","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2018-12-14 09:21:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6e5854ddeca8e24ec0219f58aa6262ccb1c70c1d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2018-12-14 10:10:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/623510/2 : FAILURE in 48m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c2874db4b15b4be59b41c7f21c28e828c99b2ed3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2018-12-14 10:33:38.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/10/623510/2/check/tempest-dsvm-full-xenial/e9fc580/ : SUCCESS in 1h 10m 40s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/10/623510/2/check/tempest-dsvm-full-xenial-py3/96b001e/ : FAILURE in 1h 02m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e80fa4e058671bbb526710b917d20a70c034e6b6","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2018-12-14 10:36:56.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/623510/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5da9846 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5435fb34f24fd62af6acace7a815d4192141e77d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2018-12-14 11:24:48.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/10/623510/2/check/nova-out-of-tree-pvm/5228396 : SUCCESS in 2h 02m 50s\n- nova-in-tree-pvm http://184.172.12.213/10/623510/2/check/nova-in-tree-pvm/ed69a91 : FAILURE in 1h 20m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d60d4622de0220bf926d7bc534265614fcb1d7bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 11:52:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/10/623510/2/check/openstack-tox-pep8/63f8394/ : SUCCESS in 7m 37s\n- openstack-tox-py27 http://logs.openstack.org/10/623510/2/check/openstack-tox-py27/03d9520/ : SUCCESS in 10m 43s\n- openstack-tox-py35 http://logs.openstack.org/10/623510/2/check/openstack-tox-py35/2cfe4bd/ : SUCCESS in 12m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/10/623510/2/check/build-openstack-sphinx-docs/65795df/html/ : SUCCESS in 6m 14s\n- tempest-full http://logs.openstack.org/10/623510/2/check/tempest-full/f650a64/ : TIMED_OUT in 2h 08m 57s\n- neutron-grenade http://logs.openstack.org/10/623510/2/check/neutron-grenade/8360389/ : SUCCESS in 48m 03s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/10/623510/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/93d5144/ : SUCCESS in 1h 02m 31s (non-voting)\n- nova-live-migration http://logs.openstack.org/10/623510/2/check/nova-live-migration/f604c3c/ : SUCCESS in 39m 51s\n- nova-tox-functional http://logs.openstack.org/10/623510/2/check/nova-tox-functional/0f9bc36/ : SUCCESS in 17m 11s\n- nova-tox-functional-py35 http://logs.openstack.org/10/623510/2/check/nova-tox-functional-py35/82db2ce/ : SUCCESS in 17m 03s\n- neutron-grenade-multinode http://logs.openstack.org/10/623510/2/check/neutron-grenade-multinode/35fb602/ : SUCCESS in 55m 19s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/10/623510/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b59f67b/ : SUCCESS in 54m 55s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/10/623510/2/check/devstack-plugin-ceph-tempest/caa1dc7/ : SUCCESS in 1h 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80e104ae0596a2b3f400e8551ff1bd7031953996","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-12-14 14:10:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm, queens and rocky are both approved.","accounts_in_message":[],"_revision_number":2},{"id":"03f5bd5f7bf7579a23934989f95064d53dad5733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 16:04:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/623510/2/check/openstack-tox-pep8/770fa1c/ : SUCCESS in 8m 06s\n- openstack-tox-py27 http://logs.openstack.org/10/623510/2/check/openstack-tox-py27/87f2da1/ : SUCCESS in 12m 02s\n- openstack-tox-py35 http://logs.openstack.org/10/623510/2/check/openstack-tox-py35/30dd492/ : SUCCESS in 13m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/10/623510/2/check/build-openstack-sphinx-docs/f44aa24/html/ : SUCCESS in 6m 19s\n- tempest-full http://logs.openstack.org/10/623510/2/check/tempest-full/ae44550/ : SUCCESS in 1h 45m 02s\n- neutron-grenade http://logs.openstack.org/10/623510/2/check/neutron-grenade/e3a69e0/ : SUCCESS in 54m 51s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/10/623510/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/cdbdd90/ : SUCCESS in 48m 23s (non-voting)\n- nova-live-migration http://logs.openstack.org/10/623510/2/check/nova-live-migration/669ac6f/ : SUCCESS in 46m 35s\n- nova-tox-functional http://logs.openstack.org/10/623510/2/check/nova-tox-functional/77bf9ca/ : SUCCESS in 17m 14s\n- nova-tox-functional-py35 http://logs.openstack.org/10/623510/2/check/nova-tox-functional-py35/1a33b1f/ : SUCCESS in 17m 14s\n- neutron-grenade-multinode http://logs.openstack.org/10/623510/2/check/neutron-grenade-multinode/449f4fb/ : SUCCESS in 50m 48s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/10/623510/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/8fc9d6b/ : SUCCESS in 49m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/10/623510/2/check/devstack-plugin-ceph-tempest/1f491e0/ : SUCCESS in 1h 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17ce5d4d9d5f1f7417ada6e99f03befa255edc03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 16:08:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a042d3144f42518de4f3fc12c316d7a59e249942","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 20:35:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/623510/2/gate/openstack-tox-pep8/bf63e54/ : SUCCESS in 8m 54s\n- openstack-tox-py27 http://logs.openstack.org/10/623510/2/gate/openstack-tox-py27/03e3071/ : SUCCESS in 13m 39s\n- openstack-tox-py35 http://logs.openstack.org/10/623510/2/gate/openstack-tox-py35/788bc66/ : SUCCESS in 16m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/10/623510/2/gate/build-openstack-sphinx-docs/4919cfb/html/ : SUCCESS in 6m 18s\n- tempest-full http://logs.openstack.org/10/623510/2/gate/tempest-full/a1eebd4/ : SUCCESS in 1h 38m 56s\n- neutron-grenade http://logs.openstack.org/10/623510/2/gate/neutron-grenade/3e88458/ : SUCCESS in 48m 27s\n- nova-live-migration http://logs.openstack.org/10/623510/2/gate/nova-live-migration/9cd36b5/ : SUCCESS in 50m 02s\n- nova-tox-functional http://logs.openstack.org/10/623510/2/gate/nova-tox-functional/b7588ad/ : SUCCESS in 18m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/10/623510/2/gate/nova-tox-functional-py35/26c19a6/ : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":2},{"id":"cadefee19923638c449abdbc77b1b051afbcb75d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 20:35:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44e2ecd8e1b5b11e3ff06344a97032b7da510624","revisions":{"c38e640f3ce9a26c966a420bf19b5feaedddd723":{"kind":"REWORK","_number":1,"created":"2018-12-07 14:52:27.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/10/623510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/623510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/623510/1"}}},"commit":{"parents":[{"commit":"c5cd6c8e50201dfaa192cdfbbc64d714eeadcd95","subject":"Update docs for _destroy_evacuated_instances","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c5cd6c8e50201dfaa192cdfbbc64d714eeadcd95"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2018-12-03 19:39:51.000000000","tz":-300},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2018-12-07 14:52:11.000000000","tz":0},"subject":"Add secret\u003dtrue to fixed_key configuration parameter","message":"Add secret\u003dtrue to fixed_key configuration parameter\n\nAdd secret\u003dtrue to fixed_key configuration parameter as that value\nshouldn\u0027t be logged.\n\nChange-Id: Ie6da21e8680b2deb6b1da3add31cd725ba855c1c\nCloses-Bug: #1806471\n(cherry picked from commit 37a036672e459b0d83b7b91120c8ec40e3759190)\n(cherry picked from commit f6b8149f512fd51bb9730ee5964cebaea331c54c)\n(cherry picked from commit 0932d957f68d3983add9ae2a39131a69ddfb1b19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c38e640f3ce9a26c966a420bf19b5feaedddd723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c38e640f3ce9a26c966a420bf19b5feaedddd723"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"c5cd6c8e50201dfaa192cdfbbc64d714eeadcd95","is_merged_in_target_branch":true,"change_id":"Ia0c67aeffe41e133de9dee5080be9c5ec1739aa6","change_number":621203,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/pike"},"44e2ecd8e1b5b11e3ff06344a97032b7da510624":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2018-12-14 09:21:18.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/10/623510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/623510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/623510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/623510/2"}}},"commit":{"parents":[{"commit":"f25cc4ac927d3879eb8bc3d66c8d1b624a10cdf8","subject":"Merge \"Make supports_direct_io work on 4096b sector size\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f25cc4ac927d3879eb8bc3d66c8d1b624a10cdf8"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2018-12-03 19:39:51.000000000","tz":-300},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2018-12-14 09:21:02.000000000","tz":0},"subject":"Add secret\u003dtrue to fixed_key configuration parameter","message":"Add secret\u003dtrue to fixed_key configuration parameter\n\nAdd secret\u003dtrue to fixed_key configuration parameter as that value\nshouldn\u0027t be logged.\n\nChange-Id: Ie6da21e8680b2deb6b1da3add31cd725ba855c1c\nCloses-Bug: #1806471\n(cherry picked from commit 37a036672e459b0d83b7b91120c8ec40e3759190)\n(cherry picked from commit 7ef3304b12e2a6a89cfedc99e09b214ae6de3d7a)\n(cherry picked from commit 33dc9f7bbd4ceecf2a590a30e6df034a32a31597)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44e2ecd8e1b5b11e3ff06344a97032b7da510624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44e2ecd8e1b5b11e3ff06344a97032b7da510624"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"f25cc4ac927d3879eb8bc3d66c8d1b624a10cdf8","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
