)]}'
{"id":"openstack%2Fnova~633560","triplet_id":"openstack%2Fnova~master~I791ac7dc1630cebbac6abca9d31dc09b3787bc31","project":"openstack/nova","branch":"master","topic":"server-create-handler-refactor","hashtags":[],"change_id":"I791ac7dc1630cebbac6abca9d31dc09b3787bc31","subject":"Cleanup return_reservation_id in ServersController.create","status":"MERGED","created":"2019-01-28 17:08:27.000000000","updated":"2019-02-23 20:36:38.000000000","submitted":"2019-02-23 20:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633560-1550954198972-d01ea473","meta_rev_id":"5653185a011fac19edd737cd88f226d6519e5fc1","_number":633560,"virtual_id_number":633560,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-02-11 02:02:25.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-10 23:18:51.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-11 00:25:48.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-23 18:04:53.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-23 17:03:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-23 16:21:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-02-23 20:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"date":"2019-02-23 16:46:11.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-23 19:05:17.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-23 18:04:30.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-02-23 15:47:30.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-14 08:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":1,"date":"2019-02-10 20:51:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-02-23 01:34:57.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-02-10 20:51:28.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-02-23 01:34:57.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-29 02:40:56.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-01-29 02:50:24.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2019-02-10 23:18:51.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-11 00:25:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-11 02:02:25.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-14 08:09:06.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2019-02-23 01:34:57.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-02-23 15:47:30.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-23 16:21:40.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-02-23 16:46:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-23 17:03:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-23 18:04:30.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-02-23 18:04:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-23 19:05:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-23 20:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"597bc723d091c97d6609396204ec82a4673da6a2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-28 17:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3d7d5d7e6e865e57351f07eaaac72738f73da0b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-01-28 17:09:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9f909b5225b3703d808b49927ca2141c0b4bc7b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-01-28 17:10:09.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/633560/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9fb23ea : FAILURE in 22s","accounts_in_message":[],"_revision_number":1},{"id":"de61b2aa3a1bfff671cd9be7edd7a65489107fc6","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-28 17:10:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40b88604ed1dd820f6e6a3c561a233653405062a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-01-28 17:16:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15eb066a4d9c35b07d6c428755223307dc8ff8c7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-01-28 17:17:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89243739d9c3a2316286c3fe2c96aff96223faaf","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-01-28 17:18:55.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/633560/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1697db3 : FAILURE in 25s","accounts_in_message":[],"_revision_number":2},{"id":"514bcf9c07a2efef07a254a4e9d66fd1a69a842b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-28 18:32:52.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial/97d6f8e/ : SUCCESS in 1h 11m 11s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial-py3/ab3e8c1/ : SUCCESS in 1h 14m 50s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/grenade-dsvm-xenial/279116f/ : SUCCESS in 1h 02m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ebd47de16f89f11b65d2140168c50646c3fa8f2a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-28 18:49:17.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/2/check-vote/ext-nova-zuul/3c0e460 : FAILURE in 39m 08s","accounts_in_message":[],"_revision_number":2},{"id":"df85f36bc411d54339f0a2d3b3bbbd7a2b263ea9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-01-28 20:31:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/60/633560/2/check/nova-out-of-tree-pvm/ccad89d : SUCCESS in 1h 27m 39s\n- nova-in-tree-pvm http://184.172.12.213/60/633560/2/check/nova-in-tree-pvm/06e857a : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0dda94681d347c1e96898ebce9043c1733479d03","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-01-28 20:53:46.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27190 : FAILURE in 19m 41s","accounts_in_message":[],"_revision_number":2},{"id":"209f66938dcc29ffd2a063612f944fb1706eedaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-28 21:46:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/60/633560/2/check/tempest-full/745ff73/ : TIMED_OUT in 2h 05m 57s\n- neutron-grenade http://logs.openstack.org/60/633560/2/check/neutron-grenade/7348f42/ : SUCCESS in 1h 01m 16s\n- grenade-py3 http://logs.openstack.org/60/633560/2/check/grenade-py3/d7c6d00/ : SUCCESS in 1h 38m 00s\n- tempest-full-py3 http://logs.openstack.org/60/633560/2/check/tempest-full-py3/ec3c8f3/ : TIMED_OUT in 2h 04m 36s\n- openstack-tox-cover http://logs.openstack.org/60/633560/2/check/openstack-tox-cover/02da8da/cover/ : SUCCESS in 18m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/2/check/openstack-tox-lower-constraints/883c20a/ : SUCCESS in 12m 55s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/2/check/openstack-tox-pep8/01d1416/ : SUCCESS in 10m 22s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/2/check/openstack-tox-py27/f342f1f/ : SUCCESS in 12m 45s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/2/check/openstack-tox-py35/c7b79c9/ : FAILURE in 12m 56s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/2/check/openstack-tox-py36/46fa540/ : SUCCESS in 12m 36s\n- openstack-tox-docs http://logs.openstack.org/60/633560/2/check/openstack-tox-docs/10095dd/html/ : SUCCESS in 6m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c0ed0b6/ : SUCCESS in 55m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/2/check/devstack-plugin-ceph-tempest/3ea35ac/ : SUCCESS in 1h 57m 30s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/6401156/ : FAILURE in 1h 24m 39s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/2/check/neutron-grenade-multinode/45772a0/ : SUCCESS in 2h 02m 10s\n- nova-live-migration http://logs.openstack.org/60/633560/2/check/nova-live-migration/ab1de98/ : SUCCESS in 1h 50m 50s\n- nova-multiattach http://logs.openstack.org/60/633560/2/check/nova-multiattach/1f03b9b/ : SUCCESS in 1h 13m 08s\n- nova-next http://logs.openstack.org/60/633560/2/check/nova-next/6ff6159/ : SUCCESS in 1h 45m 20s\n- nova-tox-functional http://logs.openstack.org/60/633560/2/check/nova-tox-functional/1188191/ : SUCCESS in 17m 04s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/2/check/nova-tox-functional-py35/26ee280/ : SUCCESS in 17m 32s\n- tempest-multinode-full http://logs.openstack.org/60/633560/2/check/tempest-multinode-full/31a6fcd/ : SUCCESS in 2h 08m 42s (non-voting)\n- tempest-slow http://logs.openstack.org/60/633560/2/check/tempest-slow/fa4edb6/ : SUCCESS in 2h 26m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9e5151f2fbe0018c79a4fbb38e2bd0deed1af2b8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-01-28 23:56:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/633560/2 : SUCCESS in 2h 29m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5425d80098c32d172e6708cd108abd082299f487","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-01-29 02:40:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"605e5d1538524cc4af365db450c0726640dccea0","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2019-01-29 02:50:24.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b723864d328342f496375a49bd36ffba173a1ce","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-01-29 03:50:12.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27204 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1c52cf2bb60ae222a457381892b9722de4805efd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-01-29 04:01:23.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/2/check-vote/ext-nova-zuul/973a4a4 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":2},{"id":"22ce2cf1f572cd8c34e871ce98eec9da2313865c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-29 04:09:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial/4351594/ : SUCCESS in 1h 09m 25s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial-py3/e0fad1f/ : SUCCESS in 1h 19m 00s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/grenade-dsvm-xenial/e8998fc/ : SUCCESS in 54m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a4f86cd93af275c5fb14047437efaec856ebfa8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-01-29 05:04:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/633560/2 : SUCCESS in 2h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"bbc458969fa6aaadfbce5102a8841ec4a1cb6945","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 05:08:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/60/633560/2/check/tempest-full/19cec72/ : SUCCESS in 1h 44m 38s\n- neutron-grenade http://logs.openstack.org/60/633560/2/check/neutron-grenade/21bd1bc/ : SUCCESS in 57m 13s\n- grenade-py3 http://logs.openstack.org/60/633560/2/check/grenade-py3/c3e999b/ : SUCCESS in 56m 44s\n- tempest-full-py3 http://logs.openstack.org/60/633560/2/check/tempest-full-py3/7dfaa4b/ : SUCCESS in 1h 32m 53s\n- openstack-tox-cover http://logs.openstack.org/60/633560/2/check/openstack-tox-cover/c94a96b/cover/ : SUCCESS in 16m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/2/check/openstack-tox-lower-constraints/efa50d7/ : SUCCESS in 13m 19s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/2/check/openstack-tox-pep8/f29092c/ : SUCCESS in 9m 44s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/2/check/openstack-tox-py27/fe8fa05/ : SUCCESS in 11m 40s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/2/check/openstack-tox-py35/941be57/ : SUCCESS in 18m 03s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/2/check/openstack-tox-py36/b0928b0/ : SUCCESS in 17m 12s\n- openstack-tox-docs http://logs.openstack.org/60/633560/2/check/openstack-tox-docs/a44094c/html/ : SUCCESS in 6m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2549e16/ : SUCCESS in 58m 43s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/2/check/devstack-plugin-ceph-tempest/3b0b32e/ : SUCCESS in 2h 02m 47s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/9fdab59/ : FAILURE in 1h 11m 19s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/2/check/neutron-grenade-multinode/091334b/ : SUCCESS in 1h 06m 48s\n- nova-live-migration http://logs.openstack.org/60/633560/2/check/nova-live-migration/3719228/ : SUCCESS in 50m 59s\n- nova-multiattach http://logs.openstack.org/60/633560/2/check/nova-multiattach/16d9b8f/ : SUCCESS in 1h 03m 21s\n- nova-next http://logs.openstack.org/60/633560/2/check/nova-next/f99f427/ : SUCCESS in 2h 09m 11s\n- nova-tox-functional http://logs.openstack.org/60/633560/2/check/nova-tox-functional/3af613a/ : SUCCESS in 19m 36s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/2/check/nova-tox-functional-py35/43cf2b5/ : SUCCESS in 16m 40s\n- tempest-multinode-full http://logs.openstack.org/60/633560/2/check/tempest-multinode-full/73ddc74/ : SUCCESS in 2h 04m 24s (non-voting)\n- tempest-slow http://logs.openstack.org/60/633560/2/check/tempest-slow/a9fb26d/ : FAILURE in 2h 06m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1c769ef719e9ebea1fd7095764c28ac165778479","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-01-29 08:48:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/60/633560/2/check/nova-out-of-tree-pvm/b05b6cf : SUCCESS in 1h 15m 16s\n- nova-in-tree-pvm http://184.172.12.213/60/633560/2/check/nova-in-tree-pvm/23a62bb : SUCCESS in 2h 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"980bba2adfaab6855dda496afb32b7759f664011","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-08 23:07:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eff2252b0d6be01817350e199a421082b7b30230","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-08 23:15:03.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/2/check-vote/ext-nova-zuul/70e9ca2 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f5a6bceeaf7d9978e11173e0395d353d2f33f006","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-09 00:20:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial/3e8c49f/ : SUCCESS in 1h 12m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/tempest-dsvm-full-xenial-py3/4bf1cc0/ : SUCCESS in 1h 09m 29s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/2/check/grenade-dsvm-xenial/c404f34/ : SUCCESS in 57m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4aa278cd778a56d9a9499706636b30429d5f36f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-09 01:25:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/60/633560/2/check/tempest-full/7956e9f/ : SUCCESS in 2h 01m 37s\n- neutron-grenade http://logs.openstack.org/60/633560/2/check/neutron-grenade/e667faa/ : SUCCESS in 55m 17s\n- grenade-py3 http://logs.openstack.org/60/633560/2/check/grenade-py3/436f0a5/ : SUCCESS in 1h 00m 22s\n- tempest-full-py3 http://logs.openstack.org/60/633560/2/check/tempest-full-py3/0ceb3a5/ : SUCCESS in 1h 37m 04s\n- openstack-tox-cover http://logs.openstack.org/60/633560/2/check/openstack-tox-cover/45417b0/cover/ : SUCCESS in 20m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/2/check/openstack-tox-lower-constraints/0038d1a/ : SUCCESS in 12m 29s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/2/check/openstack-tox-pep8/dd67acc/ : SUCCESS in 9m 26s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/2/check/openstack-tox-py27/d44999f/ : SUCCESS in 16m 22s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/2/check/openstack-tox-py35/ec7a498/ : SUCCESS in 13m 35s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/2/check/openstack-tox-py36/55c4be1/ : SUCCESS in 11m 58s\n- openstack-tox-docs http://logs.openstack.org/60/633560/2/check/openstack-tox-docs/1f9e823/html/ : SUCCESS in 6m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d89a192/ : SUCCESS in 51m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/2/check/devstack-plugin-ceph-tempest/2456641/ : SUCCESS in 1h 58m 36s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/8808617/ : FAILURE in 1h 00m 32s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/2/check/neutron-grenade-multinode/ce1cb44/ : SUCCESS in 1h 10m 09s\n- nova-live-migration http://logs.openstack.org/60/633560/2/check/nova-live-migration/84238a5/ : SUCCESS in 44m 33s\n- nova-multiattach http://logs.openstack.org/60/633560/2/check/nova-multiattach/de70eb5/ : SUCCESS in 1h 09m 00s\n- nova-next http://logs.openstack.org/60/633560/2/check/nova-next/1eaa42b/ : SUCCESS in 2h 01m 03s\n- nova-tox-functional http://logs.openstack.org/60/633560/2/check/nova-tox-functional/0d97502/ : SUCCESS in 19m 00s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/2/check/nova-tox-functional-py35/a53a478/ : SUCCESS in 16m 11s\n- tempest-multinode-full http://logs.openstack.org/60/633560/2/check/tempest-multinode-full/3a181f5/ : SUCCESS in 2h 04m 30s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/60/633560/2/check/tempest-slow-py3/c9fd90c/ : SUCCESS in 2h 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2854cc814ec3b067b6751bd2147c71934153c91a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-09 01:26:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/60/633560/2/check/EMC_VxFlexOS_NOVA/b915829/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"8148c9e569aa8308165b7ae9f4022ca1a55fb471","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-09 02:40:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/633560/2 : SUCCESS in 2h 23m 01s","accounts_in_message":[],"_revision_number":2},{"id":"20431593b475d6bb3ddf81645b9396b13ddf58a4","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-02-09 04:55:01.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/60/633560/2/check/nova-out-of-tree-pvm/7a75c52 : FAILURE in 5h 03m 52s\n- nova-in-tree-pvm http://184.172.12.213/60/633560/2/check/nova-in-tree-pvm/23a27f3 : FAILURE in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a87c523b091bcc93b168020a04e23e60c3fef403","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-10 20:51:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7c7b128b4301a02d4ef258bcd364238bcd05a3fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-10 20:53:04.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/3/check-vote/ext-nova-zuul/d1b1303 : FAILURE in 40s","accounts_in_message":[],"_revision_number":3},{"id":"9d6b56cf6fb7a6a023eb88508658942acf68d515","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-10 22:12:08.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial/e280581/ : SUCCESS in 1h 12m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial-py3/aa8d1eb/ : SUCCESS in 1h 19m 16s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/grenade-dsvm-xenial/40efaca/ : SUCCESS in 1h 00m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9e6ad1ae33a22c294470537c9ef8dfc56bb76fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-10 22:49:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/60/633560/3/check/tempest-full/83fab72/ : SUCCESS in 1h 29m 51s\n- neutron-grenade http://logs.openstack.org/60/633560/3/check/neutron-grenade/a9ef892/ : SUCCESS in 58m 17s\n- grenade-py3 http://logs.openstack.org/60/633560/3/check/grenade-py3/1c56dc9/ : SUCCESS in 59m 56s\n- tempest-full-py3 http://logs.openstack.org/60/633560/3/check/tempest-full-py3/044b175/ : SUCCESS in 1h 22m 15s\n- openstack-tox-cover http://logs.openstack.org/60/633560/3/check/openstack-tox-cover/f6f7e6f/cover/ : SUCCESS in 16m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/3/check/openstack-tox-lower-constraints/1e04043/ : SUCCESS in 14m 47s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/3/check/openstack-tox-pep8/6b99a91/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/3/check/openstack-tox-py27/86d330d/ : SUCCESS in 11m 24s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/3/check/openstack-tox-py35/df22a23/ : SUCCESS in 13m 58s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/3/check/openstack-tox-py36/1367d43/ : SUCCESS in 13m 50s\n- openstack-tox-docs http://logs.openstack.org/60/633560/3/check/openstack-tox-docs/90eb000/html/ : SUCCESS in 6m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0521f46/ : SUCCESS in 55m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/3/check/devstack-plugin-ceph-tempest/04a7e05/ : SUCCESS in 1h 53m 53s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/cbdcfad/ : FAILURE in 1h 06m 34s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/3/check/neutron-grenade-multinode/666c302/ : SUCCESS in 1h 05m 06s\n- nova-live-migration http://logs.openstack.org/60/633560/3/check/nova-live-migration/fca2726/ : SUCCESS in 43m 39s\n- nova-next http://logs.openstack.org/60/633560/3/check/nova-next/5e675ab/ : SUCCESS in 1h 34m 23s\n- nova-tox-functional http://logs.openstack.org/60/633560/3/check/nova-tox-functional/1cab11c/ : SUCCESS in 18m 03s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/3/check/nova-tox-functional-py35/9074565/ : SUCCESS in 21m 08s\n- tempest-multinode-full http://logs.openstack.org/60/633560/3/check/tempest-multinode-full/7e74534/ : SUCCESS in 1h 46m 25s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/60/633560/3/check/tempest-slow-py3/ff8e9d8/ : SUCCESS in 1h 48m 12s","accounts_in_message":[],"_revision_number":3},{"id":"8fe1aa4c86c78211340fd01c5a08919e5ed5983c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-10 22:50:18.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/60/633560/3/check/EMC_VxFlexOS_NOVA/6618fff/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"87afd167da13be08eb158849c3594b47af835652","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-10 23:18:51.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/633560/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/da6fe8b : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":3},{"id":"762f328db6d1ebe684a514d57bb959257777a7cb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-10 23:24:10.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/633560/3 : FAILURE in 2h 31m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ba15eda3dfeada23c9edbc34f2b4c6d58fa44906","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-02-11 00:25:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/60/633560/3/check-nova/Nova-ML2-Sriov/8e53046 : SUCCESS in 1h 21m 21s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/60/633560/3/check-nova/Nova-MACVTAP-ML2-Sriov/0ea187a : SUCCESS in 42m 49s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":3},{"id":"505bb266159a50dc056582ab28f1fc933c93a2d8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-02-11 02:02:25.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/60/633560/3/check/nova-out-of-tree-pvm/39560f3 : FAILURE in 5h 04m 08s\n- nova-in-tree-pvm http://184.172.12.213/60/633560/3/check/nova-in-tree-pvm/f3999e4 : FAILURE in 5h 00m 52s","accounts_in_message":[],"_revision_number":3},{"id":"89e2d19df47ccffe7f6c6aff9c2bbad123af3c0a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-12 13:50:20.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27412 : FAILURE in 2h 19m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1e9d0bcb51cc2d0372c512ee4d64d36701e02497","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2019-02-14 08:09:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6e8356a472b3f809959f01acac08fe5c502e07f1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-23 01:34:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dff5fdce5063d74bf7f85546c47cd6c31a605c23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 01:35:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aa4507e1ddc1df034a845cdf58a4ecb82b181634","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 06:20:42.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ab767313434e99387261b77a52677e40adc09ba7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-23 10:28:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3fcb6228ef2b7110c11da6898f3fa5f48b1ba7aa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-23 10:28:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c96b55f23541f3e44e0a0fd4f7a39f5e99e6641f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-23 11:12:34.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28144 : FAILURE in 43m 48s","accounts_in_message":[],"_revision_number":3},{"id":"efd57adf49d021583b462efba05b46b175a69030","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-23 11:28:33.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/3/check-vote/ext-nova-zuul/39d9fb7 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":3},{"id":"e2833c7b6693adee4d5843c0e2dd31d74bf30899","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-23 11:51:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial/eeadccb/ : SUCCESS in 1h 05m 41s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial-py3/64bce5c/ : SUCCESS in 1h 22m 26s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/grenade-dsvm-xenial/e10469a/ : SUCCESS in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c35195873f0565aeef7c2e2c9dc5cceb11d7a894","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-23 12:46:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/633560/3 : SUCCESS in 2h 17m 55s","accounts_in_message":[],"_revision_number":3},{"id":"38d5ffbfdaf334568f31521cc5f8b8375800b271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 12:54:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/60/633560/3/check/grenade-py3/6801819/ : SUCCESS in 1h 08m 51s\n- tempest-full-py3 http://logs.openstack.org/60/633560/3/check/tempest-full-py3/dbd9f22/ : SUCCESS in 1h 29m 58s\n- openstack-tox-cover http://logs.openstack.org/60/633560/3/check/openstack-tox-cover/84fb238/cover/ : SUCCESS in 16m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/3/check/openstack-tox-lower-constraints/aba7342/ : SUCCESS in 16m 32s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/3/check/openstack-tox-pep8/6ad8c98/ : SUCCESS in 9m 31s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/3/check/openstack-tox-py27/c088006/ : SUCCESS in 11m 50s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/3/check/openstack-tox-py35/ad1d49a/ : SUCCESS in 13m 03s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/3/check/openstack-tox-py36/09d2d28/ : SUCCESS in 18m 23s\n- openstack-tox-docs http://logs.openstack.org/60/633560/3/check/openstack-tox-docs/c8699db/html/ : SUCCESS in 5m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/754dd08/ : SUCCESS in 51m 04s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/3/check/devstack-plugin-ceph-tempest/fdb9fed/ : SUCCESS in 2h 05m 13s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/0d41fc4/ : SUCCESS in 1h 18m 42s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/3/check/neutron-grenade-multinode/3f275cd/ : SUCCESS in 1h 17m 13s\n- nova-live-migration http://logs.openstack.org/60/633560/3/check/nova-live-migration/5ead161/ : SUCCESS in 41m 28s\n- nova-next http://logs.openstack.org/60/633560/3/check/nova-next/55b705f/ : SUCCESS in 1h 52m 26s\n- nova-tox-functional http://logs.openstack.org/60/633560/3/check/nova-tox-functional/467df08/ : SUCCESS in 22m 14s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/3/check/nova-tox-functional-py35/60dc744/ : SUCCESS in 18m 26s\n- tempest-multinode-full http://logs.openstack.org/60/633560/3/check/tempest-multinode-full/554c8f1/ : SUCCESS in 1h 50m 22s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/60/633560/3/check/tempest-slow-py3/9e7ae2a/ : SUCCESS in 2h 20m 46s","accounts_in_message":[],"_revision_number":3},{"id":"cb9bb1b5011328d4f034f351202f694b9416430d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-23 12:55:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/60/633560/3/check/EMC_VxFlexOS_NOVA/2ecae42/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ff729ebee8bd2963b08520e17719df412c2c7b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 12:55:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6023b789d750e38782ecc7b06ffcf85ec673092c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-23 14:08:41.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-60-633560-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"c5bc6d183b6ba9bc43e3de973c192a37725bc118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 15:13:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/60/633560/3/gate/grenade-py3/405a491/ : SUCCESS in 1h 00m 18s\n- tempest-full-py3 http://logs.openstack.org/60/633560/3/gate/tempest-full-py3/45dcb37/ : FAILURE in 1h 38m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/3/gate/openstack-tox-lower-constraints/5624c90/ : SUCCESS in 11m 48s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/3/gate/openstack-tox-pep8/43a093f/ : SUCCESS in 10m 28s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py27/fa8eff3/ : SUCCESS in 11m 44s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py35/6da1bbe/ : SUCCESS in 11m 50s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py36/f8ed396/ : SUCCESS in 11m 36s\n- openstack-tox-docs http://logs.openstack.org/60/633560/3/gate/openstack-tox-docs/4a9f322/html/ : SUCCESS in 5m 59s\n- nova-live-migration http://logs.openstack.org/60/633560/3/gate/nova-live-migration/a5a6834/ : SUCCESS in 45m 50s\n- nova-tox-functional http://logs.openstack.org/60/633560/3/gate/nova-tox-functional/fac40f8/ : SUCCESS in 17m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/3/gate/nova-tox-functional-py35/1cc02b0/ : SUCCESS in 17m 46s\n- nova-next http://logs.openstack.org/60/633560/3/gate/nova-next/535b892/ : SUCCESS in 1h 56m 07s\n- tempest-slow-py3 http://logs.openstack.org/60/633560/3/gate/tempest-slow-py3/74240b3/ : SUCCESS in 2h 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fe6aeaf10a0e45e4df231a3438b50ae652900a26","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-23 15:47:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9bb0ff4bfb0628f530e5c46fa3f91b68dd5c012d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-23 15:47:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eab5ea89377eaf7d4824abedacb8eeddccc9ea97","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-23 16:21:40.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28150 : FAILURE in 34m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5788284143d1f30cc01bbf2179cfee3ea62a9fa9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-23 16:46:11.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/60/633560/3/check-vote/ext-nova-zuul/b3896c3 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":3},{"id":"85943a6337495d7275b6179c4170c8c89484aefc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-23 17:03:49.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial/03e91e5/ : SUCCESS in 1h 15m 45s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/tempest-dsvm-full-xenial-py3/601790e/ : SUCCESS in 1h 16m 02s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/60/633560/3/check/grenade-dsvm-xenial/84338da/ : SUCCESS in 1h 00m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aecde01577093214ef9e0c658b5599f5f60c07e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 18:04:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/60/633560/3/check/grenade-py3/1ec914e/ : SUCCESS in 1h 00m 36s\n- tempest-full-py3 http://logs.openstack.org/60/633560/3/check/tempest-full-py3/452cbe4/ : SUCCESS in 1h 30m 21s\n- openstack-tox-cover http://logs.openstack.org/60/633560/3/check/openstack-tox-cover/900eb97/cover/ : SUCCESS in 17m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/3/check/openstack-tox-lower-constraints/917f0c3/ : SUCCESS in 11m 27s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/3/check/openstack-tox-pep8/cb0968f/ : SUCCESS in 9m 32s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/3/check/openstack-tox-py27/070d075/ : SUCCESS in 11m 40s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/3/check/openstack-tox-py35/b06636d/ : SUCCESS in 12m 27s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/3/check/openstack-tox-py36/a60c1ed/ : SUCCESS in 13m 11s\n- openstack-tox-docs http://logs.openstack.org/60/633560/3/check/openstack-tox-docs/f41d06b/html/ : SUCCESS in 5m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/60/633560/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/de19874/ : SUCCESS in 48m 41s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/60/633560/3/check/devstack-plugin-ceph-tempest/5110a48/ : SUCCESS in 2h 00m 10s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/60/633560/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/95605c1/ : SUCCESS in 1h 03m 26s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/60/633560/3/check/neutron-grenade-multinode/b1b7d62/ : SUCCESS in 1h 08m 24s\n- nova-live-migration http://logs.openstack.org/60/633560/3/check/nova-live-migration/6a1ab17/ : SUCCESS in 49m 11s\n- nova-next http://logs.openstack.org/60/633560/3/check/nova-next/f04d5b5/ : SUCCESS in 1h 41m 00s\n- nova-tox-functional http://logs.openstack.org/60/633560/3/check/nova-tox-functional/8ed12c1/ : SUCCESS in 18m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/3/check/nova-tox-functional-py35/04d6f07/ : SUCCESS in 17m 38s\n- tempest-multinode-full http://logs.openstack.org/60/633560/3/check/tempest-multinode-full/ffea790/ : SUCCESS in 1h 50m 21s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/60/633560/3/check/tempest-slow-py3/1054c09/ : SUCCESS in 2h 11m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6dfced681fa7dbfadeb806aeb8b07181794d38c6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-23 18:04:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/633560/3 : SUCCESS in 2h 16m 40s","accounts_in_message":[],"_revision_number":3},{"id":"435c6a1ad531595e76d559314fef518285d7e410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 18:04:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7dcb65869bcb85db39866eb53cf5845ee02cc6a7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-23 18:04:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/60/633560/3/check/EMC_VxFlexOS_NOVA/1c07ff8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"fd3ed441703a98966e1c1ce571360c217ada6167","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-23 19:05:17.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-60-633560-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"417d27bca11f48db9f33ce9e6259451887812f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 20:36:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/60/633560/3/gate/grenade-py3/5b4e363/ : SUCCESS in 1h 00m 52s\n- tempest-full-py3 http://logs.openstack.org/60/633560/3/gate/tempest-full-py3/0c4ec15/ : SUCCESS in 1h 30m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/633560/3/gate/openstack-tox-lower-constraints/3d85a66/ : SUCCESS in 13m 34s\n- openstack-tox-pep8 http://logs.openstack.org/60/633560/3/gate/openstack-tox-pep8/f6f630b/ : SUCCESS in 9m 18s\n- openstack-tox-py27 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py27/9f19668/ : SUCCESS in 11m 59s\n- openstack-tox-py35 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py35/7290522/ : SUCCESS in 12m 22s\n- openstack-tox-py36 http://logs.openstack.org/60/633560/3/gate/openstack-tox-py36/f09eb2e/ : SUCCESS in 15m 25s\n- openstack-tox-docs http://logs.openstack.org/60/633560/3/gate/openstack-tox-docs/d04d9c9/html/ : SUCCESS in 5m 54s\n- nova-live-migration http://logs.openstack.org/60/633560/3/gate/nova-live-migration/1cc7085/ : SUCCESS in 46m 22s\n- nova-tox-functional http://logs.openstack.org/60/633560/3/gate/nova-tox-functional/49a21d2/ : SUCCESS in 17m 22s\n- nova-tox-functional-py35 http://logs.openstack.org/60/633560/3/gate/nova-tox-functional-py35/d39336b/ : SUCCESS in 16m 40s\n- nova-next http://logs.openstack.org/60/633560/3/gate/nova-next/1c04c63/ : SUCCESS in 2h 23m 05s\n- tempest-slow-py3 http://logs.openstack.org/60/633560/3/gate/tempest-slow-py3/ef4383b/ : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a6c92d5683f7e7ccaed32690fa14a4e87a544ece","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 20:36:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c0cd8fe8c1cb217063eacfa8aa3d4d46093c71a0","revisions":{"0489c0f70a5e2a111d74c9f81cd1a3b6d2bc934e":{"kind":"REWORK","_number":1,"created":"2019-01-28 17:08:27.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/60/633560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/633560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/633560/1"}}},"commit":{"parents":[{"commit":"bad78b8086247bea40f62ee4755dfb442041250c","subject":"Refactor bdm handling in ServersController.create method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bad78b8086247bea40f62ee4755dfb442041250c"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-01-28 17:06:51.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-01-28 17:06:51.000000000","tz":-300},"subject":"Cleanup return_reservation_id in ServersController.create","message":"Cleanup return_reservation_id in ServersController.create\n\nThe create() method was getting the return_reservation_id\nvalue out of the request body and putting it into the\ncreate_kwargs dict just to pop it off, which is pointless.\n\nThis change simply avoids messing with the create_kwargs\ndict for return_reservation_id but keeps the comment\nabout why we don\u0027t pass it through to API.create().\n\nChange-Id: I791ac7dc1630cebbac6abca9d31dc09b3787bc31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0489c0f70a5e2a111d74c9f81cd1a3b6d2bc934e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0489c0f70a5e2a111d74c9f81cd1a3b6d2bc934e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bad78b8086247bea40f62ee4755dfb442041250c","is_merged_in_target_branch":false,"change_id":"Ice3fdbfe091e8e008ab4875a74030a61cd006193","change_number":633535,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5593b0a2a5c82635725dfac59124e9203aa2ed50":{"kind":"REWORK","_number":2,"created":"2019-01-28 17:16:59.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/60/633560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/633560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/633560/2"}}},"commit":{"parents":[{"commit":"bad78b8086247bea40f62ee4755dfb442041250c","subject":"Refactor bdm handling in ServersController.create method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bad78b8086247bea40f62ee4755dfb442041250c"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-01-28 17:06:51.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-01-28 17:16:28.000000000","tz":-300},"subject":"Cleanup return_reservation_id in ServersController.create","message":"Cleanup return_reservation_id in ServersController.create\n\nThe create() method was getting the return_reservation_id\nvalue out of the request body and putting it into the\ncreate_kwargs dict just to pop it off, which is pointless.\n\nThis change simply avoids messing with the create_kwargs\ndict for return_reservation_id and moves it to where it\nis used so the comment block is removed since it is no\nlonger helpful.\n\nChange-Id: I791ac7dc1630cebbac6abca9d31dc09b3787bc31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5593b0a2a5c82635725dfac59124e9203aa2ed50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5593b0a2a5c82635725dfac59124e9203aa2ed50"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bad78b8086247bea40f62ee4755dfb442041250c","is_merged_in_target_branch":false,"change_id":"Ice3fdbfe091e8e008ab4875a74030a61cd006193","change_number":633535,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c0cd8fe8c1cb217063eacfa8aa3d4d46093c71a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-02-10 20:51:28.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/60/633560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/633560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/633560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/633560/3"}}},"commit":{"parents":[{"commit":"487bed050455047a2a03c4f513f542340659b385","subject":"Refactor bdm handling in ServersController.create method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/487bed050455047a2a03c4f513f542340659b385"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-01-28 17:06:51.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-10 20:46:10.000000000","tz":-300},"subject":"Cleanup return_reservation_id in ServersController.create","message":"Cleanup return_reservation_id in ServersController.create\n\nThe create() method was getting the return_reservation_id\nvalue out of the request body and putting it into the\ncreate_kwargs dict just to pop it off, which is pointless.\n\nThis change simply avoids messing with the create_kwargs\ndict for return_reservation_id and moves it to where it\nis used so the comment block is removed since it is no\nlonger helpful.\n\nChange-Id: I791ac7dc1630cebbac6abca9d31dc09b3787bc31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0cd8fe8c1cb217063eacfa8aa3d4d46093c71a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0cd8fe8c1cb217063eacfa8aa3d4d46093c71a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"487bed050455047a2a03c4f513f542340659b385","is_merged_in_target_branch":true,"change_id":"Ice3fdbfe091e8e008ab4875a74030a61cd006193","change_number":633535,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
