)]}'
{"id":"openstack%2Fnova~634204","triplet_id":"openstack%2Fnova~master~I614dea09547c8f86f667b1daeb18f69fea7937fb","project":"openstack/nova","branch":"master","hashtags":[],"change_id":"I614dea09547c8f86f667b1daeb18f69fea7937fb","subject":"Switch tempest-slow to be run on python 3","status":"MERGED","created":"2019-01-31 11:15:37.000000000","updated":"2019-02-11 07:31:28.000000000","submitted":"2019-02-07 01:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"634204-1549502171903-74ae150c","meta_rev_id":"555807fcd3da8c020388b02a37083eb760ac4767","_number":634204,"virtual_id_number":634204,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-01-31 11:38:52.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-01-31 15:52:32.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-01-31 13:37:36.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-01-31 17:08:43.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-01-31 13:43:21.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-01-31 12:36:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-01-31 11:24:30.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-02-07 01:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-05 00:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-02-06 22:51:05.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-02-02 21:14:04.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-02-06 22:51:05.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-31 11:24:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-01-31 11:38:52.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-31 12:36:14.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-31 13:37:36.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-01-31 13:43:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-31 15:52:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-31 17:08:43.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-02 21:14:04.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-02-05 00:15:41.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-02-06 22:51:05.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-02-07 01:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b9fcefa22a478c63b70a0f45ab79a19df327d47","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-01-31 11:15:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b846e9d7d25994e1c899dcdca104ca2946ea978d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 11:23:41.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"c9a758045b95dce758a3dd0ebe3a2ad2e21f5a8f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-01-31 11:24:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6af1c11d9c0e90902d9e10fe9dd7a5a7332e9af","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-01-31 11:38:52.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/04/634204/1/check/nova-out-of-tree-pvm/4c5038b : FAILURE in 22m 36s\n- nova-in-tree-pvm http://184.172.12.213/04/634204/1/check/nova-in-tree-pvm/e31f098 : FAILURE in 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fe10ecb284bde7ed8b10c590a6140ff2ee5c9c3f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-01-31 12:36:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/04/634204/1/check/tempest-dsvm-full-xenial/5953733/ : SUCCESS in 1h 12m 00s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/04/634204/1/check/tempest-dsvm-full-xenial-py3/1e1bcf9/ : SUCCESS in 1h 20m 02s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/04/634204/1/check/grenade-dsvm-xenial/3a6ce2b/ : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65ca5d38ef20d658245c89420a7b2202b756bbfb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-01-31 13:37:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/634204/1 : FAILURE in 2h 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"309f4d1fd3aee13d1859a5b3ea461258e9d78ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 13:42:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/04/634204/1/check/tempest-full/7a6d583/ : SUCCESS in 1h 44m 54s\n- neutron-grenade http://logs.openstack.org/04/634204/1/check/neutron-grenade/4072e0e/ : SUCCESS in 56m 27s\n- grenade-py3 http://logs.openstack.org/04/634204/1/check/grenade-py3/619798c/ : SUCCESS in 1h 05m 39s\n- tempest-full-py3 http://logs.openstack.org/04/634204/1/check/tempest-full-py3/49cf11b/ : SUCCESS in 1h 33m 02s\n- openstack-tox-cover http://logs.openstack.org/04/634204/1/check/openstack-tox-cover/bc16932/cover/ : SUCCESS in 19m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/634204/1/check/openstack-tox-lower-constraints/df5ba57/ : SUCCESS in 14m 05s\n- openstack-tox-pep8 http://logs.openstack.org/04/634204/1/check/openstack-tox-pep8/b62df38/ : SUCCESS in 9m 42s\n- openstack-tox-py27 http://logs.openstack.org/04/634204/1/check/openstack-tox-py27/986b4bf/ : SUCCESS in 11m 43s\n- openstack-tox-py35 http://logs.openstack.org/04/634204/1/check/openstack-tox-py35/64dbc0f/ : SUCCESS in 13m 43s\n- openstack-tox-py36 http://logs.openstack.org/04/634204/1/check/openstack-tox-py36/722039b/ : SUCCESS in 16m 25s\n- openstack-tox-docs http://logs.openstack.org/04/634204/1/check/openstack-tox-docs/5d45ebf/html/ : SUCCESS in 6m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/04/634204/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/96dadc9/ : SUCCESS in 55m 13s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/04/634204/1/check/devstack-plugin-ceph-tempest/3ba3be9/ : TIMED_OUT in 2h 05m 35s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/04/634204/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/2e32b50/ : FAILURE in 1h 07m 23s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/04/634204/1/check/neutron-grenade-multinode/b0f0c49/ : SUCCESS in 1h 00m 01s\n- nova-live-migration http://logs.openstack.org/04/634204/1/check/nova-live-migration/8c38efe/ : SUCCESS in 45m 52s\n- nova-lvm http://logs.openstack.org/04/634204/1/check/nova-lvm/5473e5a/ : SUCCESS in 1h 17m 49s (non-voting)\n- nova-multiattach http://logs.openstack.org/04/634204/1/check/nova-multiattach/3bac776/ : SUCCESS in 1h 03m 56s\n- nova-next http://logs.openstack.org/04/634204/1/check/nova-next/21073fc/ : SUCCESS in 2h 04m 57s\n- nova-tox-functional http://logs.openstack.org/04/634204/1/check/nova-tox-functional/556ee25/ : SUCCESS in 19m 54s\n- nova-tox-functional-py35 http://logs.openstack.org/04/634204/1/check/nova-tox-functional-py35/db42f5d/ : SUCCESS in 18m 46s\n- tempest-multinode-full http://logs.openstack.org/04/634204/1/check/tempest-multinode-full/3ab5689/ : SUCCESS in 1h 49m 58s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/04/634204/1/check/tempest-slow-py3/94fe597/ : SUCCESS in 1h 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a58925d14358b32078d02d9fdc1a50a8f3083ccb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-01-31 13:43:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/04/634204/1/check/EMC_VxFlexOS_NOVA/0516805/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"360ca2dae48f6c44ca6355133cb6e88009a2641c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-01-31 15:52:32.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-04-634204-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"37e6cf1627b34c284c85854211dab7f827562668","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-01-31 17:08:43.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/634204/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/0107c93 : FAILURE in 1h 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a6b061ca44122779a9619d17bbb6b1bb0fafdf47","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-02 21:13:55.000000000","message":"Patch Set 1:\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"8b203bcf69abe42271f6f3c1d7d9ddccb41d1549","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-02 21:14:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d26b9ae126a3b388d8608486d0bbe249745d040","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-02-05 00:15:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"16be3eaa34bf7ce332b10676e0b2012167aa1fb0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-06 22:51:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f9fee5fc39b5aacda5dae8b6feacdced038905e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 22:51:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"904c4bd526354a0e0a3ee510a375d2aae225683d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 01:16:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/04/634204/1/gate/tempest-full/51bfa85/ : SUCCESS in 1h 43m 55s\n- neutron-grenade http://logs.openstack.org/04/634204/1/gate/neutron-grenade/7309f81/ : SUCCESS in 58m 01s\n- grenade-py3 http://logs.openstack.org/04/634204/1/gate/grenade-py3/ea5497e/ : SUCCESS in 1h 03m 39s\n- tempest-full-py3 http://logs.openstack.org/04/634204/1/gate/tempest-full-py3/315f327/ : SUCCESS in 1h 28m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/634204/1/gate/openstack-tox-lower-constraints/bc5ab6b/ : SUCCESS in 12m 23s\n- openstack-tox-pep8 http://logs.openstack.org/04/634204/1/gate/openstack-tox-pep8/e75fcad/ : SUCCESS in 10m 08s\n- openstack-tox-py27 http://logs.openstack.org/04/634204/1/gate/openstack-tox-py27/9fcab7a/ : SUCCESS in 12m 03s\n- openstack-tox-py35 http://logs.openstack.org/04/634204/1/gate/openstack-tox-py35/a0ffbfa/ : SUCCESS in 17m 48s\n- openstack-tox-py36 http://logs.openstack.org/04/634204/1/gate/openstack-tox-py36/de553c1/ : SUCCESS in 16m 57s\n- openstack-tox-docs http://logs.openstack.org/04/634204/1/gate/openstack-tox-docs/4fbe258/html/ : SUCCESS in 7m 44s\n- nova-live-migration http://logs.openstack.org/04/634204/1/gate/nova-live-migration/c66e3d2/ : SUCCESS in 49m 39s\n- nova-multiattach http://logs.openstack.org/04/634204/1/gate/nova-multiattach/a39d08d/ : SUCCESS in 1h 18m 23s\n- nova-tox-functional http://logs.openstack.org/04/634204/1/gate/nova-tox-functional/3177f88/ : SUCCESS in 20m 19s\n- nova-tox-functional-py35 http://logs.openstack.org/04/634204/1/gate/nova-tox-functional-py35/083659b/ : SUCCESS in 19m 43s\n- nova-next http://logs.openstack.org/04/634204/1/gate/nova-next/fb194d7/ : SUCCESS in 1h 56m 42s\n- tempest-slow-py3 http://logs.openstack.org/04/634204/1/gate/tempest-slow-py3/bc96b46/ : SUCCESS in 2h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"680ad4de84fbc7082daab57bf53430b50f3fc996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 01:16:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"555807fcd3da8c020388b02a37083eb760ac4767","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-11 07:31:28.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27237 : SUCCESS in 2h 34m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36efe194fe82f28b8865c62b34c7802a69337af5","revisions":{"36efe194fe82f28b8865c62b34c7802a69337af5":{"kind":"REWORK","_number":1,"created":"2019-01-31 11:15:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/634204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/634204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/634204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/634204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/634204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/634204/1"}}},"commit":{"parents":[{"commit":"1822e4f9d4ec66425134bd3b8bb6d94ff4ada901","subject":"Merge \"Fix config docs for handle_virt_lifecycle_events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1822e4f9d4ec66425134bd3b8bb6d94ff4ada901"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-01-31 11:07:55.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-01-31 11:15:34.000000000","tz":0},"subject":"Switch tempest-slow to be run on python 3","message":"Switch tempest-slow to be run on python 3\n\nThis patch switches tempest-slow to new\ntempest-slow-py3 job.\n\nFor coverage of tempest-slow tests on py2,\nTempest check pipeline will continue running\nthis job on check pipeline. As this tempest-slow\njob is not from integrated gate template, it is\nok to run only py3 job on project side.\n\nDepends-On: https://review.openstack.org/633983\nChange-Id: I614dea09547c8f86f667b1daeb18f69fea7937fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/36efe194fe82f28b8865c62b34c7802a69337af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/36efe194fe82f28b8865c62b34c7802a69337af5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1822e4f9d4ec66425134bd3b8bb6d94ff4ada901","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
