)]}'
{"id":"openstack%2Fnova~636996","triplet_id":"openstack%2Fnova~master~Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15","project":"openstack/nova","branch":"master","topic":"remove_duplicate_clean_up","hashtags":[],"change_id":"Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15","subject":"Remove duplicate cleanup in functional tests","status":"MERGED","created":"2019-02-14 15:56:07.000000000","updated":"2019-03-13 15:42:59.000000000","submitted":"2019-03-13 15:42:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636996-1552491779139-7840d4c1","meta_rev_id":"881398459d743210c5c4b8504bdaf999c4d5da64","_number":636996,"virtual_id_number":636996,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-03-08 08:55:02.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-08 21:00:57.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-03-08 10:25:35.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-03-08 01:31:36.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-03-13 15:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-12 21:58:13.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-13 08:22:14.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-03-13 08:22:14.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-14 18:42:07.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-27 05:38:08.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-07 11:20:03.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-03-08 01:31:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-08 08:55:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-08 10:25:35.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-03-08 21:00:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-12 21:58:13.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-13 08:22:14.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-03-13 15:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ddaa5e404c7bf6581d6b51f772d6c0d4f23c7f7","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-02-14 15:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4afb0df28ea18924015c5e0541436f7a09f695ce","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-14 15:56:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8a831d697dc7384a3bd19764b17741a1a727d20","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-14 17:02:54.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/636996/1/check-vote/ext-nova-zuul/11e3cfe : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"af0c6cf37cf6eee608f8317d85b279747c9adaa3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-14 17:53:36.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27643 : FAILURE in 1h 30m 34s","accounts_in_message":[],"_revision_number":1},{"id":"412c16a1910984c6251a87652981cb2e3ca3ded3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 18:19:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/636996/1/check/openstack-tox-cover/10f27f1/cover/ : SUCCESS in 16m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/636996/1/check/openstack-tox-lower-constraints/329da72/ : SUCCESS in 21m 54s\n- openstack-tox-pep8 http://logs.openstack.org/96/636996/1/check/openstack-tox-pep8/d45b681/ : SUCCESS in 10m 03s\n- openstack-tox-py27 http://logs.openstack.org/96/636996/1/check/openstack-tox-py27/bef8cd9/ : SUCCESS in 30m 36s\n- openstack-tox-py35 http://logs.openstack.org/96/636996/1/check/openstack-tox-py35/cb12e5e/ : SUCCESS in 12m 32s\n- openstack-tox-py36 http://logs.openstack.org/96/636996/1/check/openstack-tox-py36/e2c1cf2/ : SUCCESS in 11m 44s\n- openstack-tox-docs http://logs.openstack.org/96/636996/1/check/openstack-tox-docs/d3ebb50/html/ : SUCCESS in 6m 38s\n- nova-tox-functional http://logs.openstack.org/96/636996/1/check/nova-tox-functional/6db2866/ : SUCCESS in 19m 23s\n- nova-tox-functional-py35 http://logs.openstack.org/96/636996/1/check/nova-tox-functional-py35/7b10d67/ : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"83ecc749ccc49135272840d2c4b0937457ae82a8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-14 18:19:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/636996/1/check/EMC_VxFlexOS_NOVA/6252760/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c0dadbd28173a8262504bcea3e4cc328a77b09b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-02-14 18:42:07.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/636996/1/check/nova-out-of-tree-pvm/d6e6a3b : FAILURE in 2h 14m 34s\n- nova-in-tree-pvm http://184.172.12.213/96/636996/1/check/nova-in-tree-pvm/2ad736e : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":1},{"id":"de613fe7cee41934b2705cc5b6cc37afa708c9a5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-14 20:13:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/636996/1 : FAILURE in 2h 43m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c88c98859cbc6028a4104e10bbe383b770b44c44","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-14 20:39:15.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-636996-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"e9ad05779b898f8be20330000a494d5e095c7bac","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-14 21:13:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis is correct as far as it goes, but I think we could go further. AFAICT there\u0027s no reason to wait until the end of the test case to restore_nodes - we should be able to do it as soon as we\u0027re done using the fake node list to start the fake compute services.\n\nI could be talked out of this.","accounts_in_message":[],"_revision_number":1},{"id":"15ae8f770ed6839f8343ab67f34b7a518b4f9018","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-02-27 05:37:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d4f987cfb0282452ee5425865e6a2f65b432e1b","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-02-27 05:37:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1cdb79eb0b7ee971e5da16b1b665d0b7390878bd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 05:38:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae5b3ae2538019dffe98834f9468c78c6db62bfc","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-27 05:39:54.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/636996/2 : FAILURE in 1m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d79170e2cf700be8e48cf161d3e22f48e626b589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 05:59:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/636996/2/check/openstack-tox-cover/d60f818/cover/ : SUCCESS in 17m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/636996/2/check/openstack-tox-lower-constraints/142ee0b/ : SUCCESS in 14m 58s\n- openstack-tox-pep8 http://logs.openstack.org/96/636996/2/check/openstack-tox-pep8/6a86eaa/ : SUCCESS in 11m 22s\n- openstack-tox-py27 http://logs.openstack.org/96/636996/2/check/openstack-tox-py27/ede8c21/ : SUCCESS in 12m 53s\n- openstack-tox-py35 http://logs.openstack.org/96/636996/2/check/openstack-tox-py35/220d92b/ : SUCCESS in 11m 36s\n- openstack-tox-py36 http://logs.openstack.org/96/636996/2/check/openstack-tox-py36/abe1623/ : SUCCESS in 14m 16s\n- openstack-tox-docs http://logs.openstack.org/96/636996/2/check/openstack-tox-docs/06452af/html/ : SUCCESS in 5m 22s\n- nova-tox-functional http://logs.openstack.org/96/636996/2/check/nova-tox-functional/00531c0/ : SUCCESS in 18m 06s\n- nova-tox-functional-py35 http://logs.openstack.org/96/636996/2/check/nova-tox-functional-py35/a218573/ : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2bc06a31337b220c631b9130ba0b38036c1e15cd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 06:00:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/636996/2/check/EMC_VxFlexOS_NOVA/c283ce3/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"c20ff06a144800630ac9282b89a391029cd813d3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-27 06:08:22.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28311 : FAILURE in 30m 05s","accounts_in_message":[],"_revision_number":2},{"id":"23d7efd3b04fb6a62e8a87bfd9067b831c613f0a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-27 12:36:46.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-636996-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"35c86454e5fda76aefa45e2c70b1784a98a6ecfa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-27 21:46:53.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/636996/2/check-vote/ext-nova-zuul/e2a7d66 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":2},{"id":"77c44fb09812e9dfa2062c648b4a1c1361ee8b0f","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-07 10:03:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"60c3340e3a71dcf110f5b846262e560390ac665d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 10:30:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/636996/3/check/openstack-tox-cover/bfa8546/cover/ : SUCCESS in 14m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/636996/3/check/openstack-tox-lower-constraints/56b1901/ : SUCCESS in 13m 37s\n- openstack-tox-pep8 http://logs.openstack.org/96/636996/3/check/openstack-tox-pep8/5e6d03b/ : SUCCESS in 9m 55s\n- openstack-tox-py27 http://logs.openstack.org/96/636996/3/check/openstack-tox-py27/572ad10/ : SUCCESS in 18m 59s\n- openstack-tox-py35 http://logs.openstack.org/96/636996/3/check/openstack-tox-py35/b3b9cac/ : SUCCESS in 13m 44s\n- openstack-tox-py36 http://logs.openstack.org/96/636996/3/check/openstack-tox-py36/dc9d0e3/ : SUCCESS in 17m 43s\n- openstack-tox-docs http://logs.openstack.org/96/636996/3/check/openstack-tox-docs/2c830ac/html/ : SUCCESS in 7m 25s\n- nova-tox-functional http://logs.openstack.org/96/636996/3/check/nova-tox-functional/ee6ce6f/ : SUCCESS in 19m 04s\n- nova-tox-functional-py35 http://logs.openstack.org/96/636996/3/check/nova-tox-functional-py35/498a1f1/ : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":3},{"id":"bfca2a56ff58cdc28a744af014bc3d7d286a0ecc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-07 10:31:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/636996/3/check/EMC_VxFlexOS_NOVA/e12c7df/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"0c2f490cc577b0e3c5dc522559af9b53084b1380","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-07 11:20:03.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28868 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":3},{"id":"f28bf3a6b3c18de96e7406136564b8cd7edfab51","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 12:15:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/636996/3 : SUCCESS in 2h 10m 59s","accounts_in_message":[],"_revision_number":3},{"id":"58459b3be449544572e164ebb8767f8eee56a8b8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-07 18:46:28.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/636996/3/check-vote/ext-nova-zuul/decf20a : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":3},{"id":"661e6f8ab2887110be0bb189c3f86a6315322db8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-07 23:59:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ba7331b6b5a743a5068e6579891fecc6d82a224d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-08 00:18:41.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-636996-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"6cfb17ff1ab34fd13a364e821e4f9451aa331fd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 01:31:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/96/636996/4/check/openstack-tox-cover/5b9fddd/cover/ : SUCCESS in 16m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/636996/4/check/openstack-tox-lower-constraints/d610b21/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/96/636996/4/check/openstack-tox-pep8/1d50c22/ : SUCCESS in 10m 18s\n- openstack-tox-py27 http://logs.openstack.org/96/636996/4/check/openstack-tox-py27/452dc72/ : SUCCESS in 14m 14s\n- openstack-tox-py35 http://logs.openstack.org/96/636996/4/check/openstack-tox-py35/8e54bf9/ : SUCCESS in 12m 13s\n- openstack-tox-py36 http://logs.openstack.org/96/636996/4/check/openstack-tox-py36/e61b8fe/ : SUCCESS in 11m 30s\n- openstack-tox-docs http://logs.openstack.org/96/636996/4/check/openstack-tox-docs/cd49601/html/ : SUCCESS in 6m 15s\n- nova-tox-functional http://logs.openstack.org/96/636996/4/check/nova-tox-functional/1e3effb/ : SUCCESS in 19m 10s\n- nova-tox-functional-py35 http://logs.openstack.org/96/636996/4/check/nova-tox-functional-py35/7779793/ : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a2a94b7c84f4b5a8ca5800962fa5ac7d5be76cfe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 01:31:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/636996/4/check/EMC_VxFlexOS_NOVA/923424d/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"8770c051a9d36fc92e3263f575f20c5e2c387179","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-08 08:55:02.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-636996-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"b7d33e918a726a986a8c3e4b477c09c145b16986","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 10:25:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/636996/4 : SUCCESS in 2h 22m 35s","accounts_in_message":[],"_revision_number":4},{"id":"1e048b8ba75deeabd0ceed5d5a8fb71d9ddbfe60","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-08 21:00:57.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/636996/4/check-vote/ext-nova-zuul/092970e : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":4},{"id":"266e0a6a884f3cfc509dd80de5d0d66f4f21336f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-12 21:58:13.000000000","message":"Patch Set 4: Code-Review+2\n\nYeah, okay. I\u0027m actually not convinced we *ever* need to restore_nodes, or other things would be breaking mysteriously along the way. But this is better than it was.","accounts_in_message":[],"_revision_number":4},{"id":"b45469c1df7b1c418cff34a85891345cca9e38f3","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-03-13 08:22:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05e2d648fb77384a44fb54b62db68a852f3cfa32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 08:23:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f8bffa2a127d6399fd9f230cf0626ab4e34d711","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 15:42:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/96/636996/4/gate/openstack-tox-lower-constraints/ac9159f/ : SUCCESS in 14m 06s\n- openstack-tox-pep8 http://logs.openstack.org/96/636996/4/gate/openstack-tox-pep8/a925acc/ : SUCCESS in 11m 33s\n- openstack-tox-py27 http://logs.openstack.org/96/636996/4/gate/openstack-tox-py27/b98fbd8/ : SUCCESS in 15m 22s\n- openstack-tox-py35 http://logs.openstack.org/96/636996/4/gate/openstack-tox-py35/ac0cc24/ : SUCCESS in 14m 53s\n- openstack-tox-py36 http://logs.openstack.org/96/636996/4/gate/openstack-tox-py36/de410c5/ : SUCCESS in 14m 57s\n- openstack-tox-docs http://logs.openstack.org/96/636996/4/gate/openstack-tox-docs/1b314a0/html/ : SUCCESS in 8m 48s\n- nova-tox-functional http://logs.openstack.org/96/636996/4/gate/nova-tox-functional/55b65ce/ : SUCCESS in 23m 38s\n- nova-tox-functional-py35 http://logs.openstack.org/96/636996/4/gate/nova-tox-functional-py35/7082dd7/ : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":4},{"id":"54a645b1f3ae410626df8e38f7eda09bb535b57b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 15:42:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bda4ae3884cb11d0cc9b510e5cf7c46b46d6d727","revisions":{"7baa1c93099283967e158d4b2d18419f465d644b":{"kind":"REWORK","_number":1,"created":"2019-02-14 15:56:07.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/96/636996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/636996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/636996/1"}}},"commit":{"parents":[{"commit":"ee57c92e8b675adf3f35de29f7cea9c9d042d44b","subject":"Merge \"Ignore VolumeAttachmentNotFound exception in compute.manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee57c92e8b675adf3f35de29f7cea9c9d042d44b"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-14 15:27:42.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-14 15:54:48.000000000","tz":540},"subject":"Remove duplicate cleanup in functional tests","message":"Remove duplicate cleanup in functional tests\n\nIt is not necessary to call the following statement multiple times\nin some functional tests.\n\nself.addCleanup(fake.restore_nodes)\n\nSo remove duplicate calls in the tests.\n\nTrivalFix\nChange-Id: Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7baa1c93099283967e158d4b2d18419f465d644b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7baa1c93099283967e158d4b2d18419f465d644b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ee57c92e8b675adf3f35de29f7cea9c9d042d44b","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"001059717902eefb247c8ee7530a59e249002e51":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-27 05:37:45.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/96/636996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/636996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/636996/2"}}},"commit":{"parents":[{"commit":"6efa3861a5a829ba5883ff191e2552b063028bb0","subject":"Merge \"Avoid BadRequest error log on volume attachment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6efa3861a5a829ba5883ff191e2552b063028bb0"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-14 15:27:42.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-27 05:37:45.000000000","tz":0},"subject":"Remove duplicate cleanup in functional tests","message":"Remove duplicate cleanup in functional tests\n\nIt is not necessary to call the following statement multiple times\nin some functional tests.\n\nself.addCleanup(fake.restore_nodes)\n\nSo remove duplicate calls in the tests.\n\nTrivalFix\nChange-Id: Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/001059717902eefb247c8ee7530a59e249002e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/001059717902eefb247c8ee7530a59e249002e51"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6efa3861a5a829ba5883ff191e2552b063028bb0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ca3a9714f018f36c824b3effdc582b052483f02f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-07 10:03:30.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/96/636996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/636996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/636996/3"}}},"commit":{"parents":[{"commit":"221026eafeb915dbda8ed7e9a4a33c43f251817a","subject":"Merge \"fakelibvirt: Add ability to generate fake PCI devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/221026eafeb915dbda8ed7e9a4a33c43f251817a"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-14 15:27:42.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-03-07 10:03:30.000000000","tz":0},"subject":"Remove duplicate cleanup in functional tests","message":"Remove duplicate cleanup in functional tests\n\nIt is not necessary to call the following statement multiple times\nin some functional tests.\n\nself.addCleanup(fake.restore_nodes)\n\nSo remove duplicate calls in the tests.\n\nTrivalFix\nChange-Id: Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca3a9714f018f36c824b3effdc582b052483f02f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca3a9714f018f36c824b3effdc582b052483f02f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"221026eafeb915dbda8ed7e9a4a33c43f251817a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bda4ae3884cb11d0cc9b510e5cf7c46b46d6d727":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-03-07 23:59:06.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/96/636996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/636996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/636996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/636996/4"}}},"commit":{"parents":[{"commit":"221026eafeb915dbda8ed7e9a4a33c43f251817a","subject":"Merge \"fakelibvirt: Add ability to generate fake PCI devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/221026eafeb915dbda8ed7e9a4a33c43f251817a"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-02-14 15:27:42.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-03-07 23:58:58.000000000","tz":0},"subject":"Remove duplicate cleanup in functional tests","message":"Remove duplicate cleanup in functional tests\n\nIt is not necessary to call the following statement multiple times in\nsome functional tests.\n\nself.addCleanup(fake.restore_nodes)\n\nSo remove duplicate calls in the tests.\n\nTrivalFix\nChange-Id: Iaae6fc4a66145576f4a4fc1cea452ef6acbadb15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bda4ae3884cb11d0cc9b510e5cf7c46b46d6d727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bda4ae3884cb11d0cc9b510e5cf7c46b46d6d727"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"221026eafeb915dbda8ed7e9a4a33c43f251817a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
