)]}'
{"id":"openstack%2Fnova~637965","triplet_id":"openstack%2Fnova~master~I2773cfb5b3c2a13d514899c64dc052086134a5c4","project":"openstack/nova","branch":"master","topic":"xml_matchers_fix","hashtags":[],"change_id":"I2773cfb5b3c2a13d514899c64dc052086134a5c4","subject":"Don\u0027t reference unassigned variable in xml matchers in test","status":"ABANDONED","created":"2019-02-19 18:40:23.000000000","updated":"2019-03-01 17:21:56.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"910a85c01049a512acd6da792fa19643d791044e","_number":637965,"virtual_id_number":637965,"owner":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-02-19 22:08:49.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2019-02-19 21:47:44.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"date":"2019-02-19 21:21:05.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"date":"2019-02-19 21:10:10.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-02-19 22:22:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":-1,"date":"2019-02-19 22:26:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":-1,"date":"2019-02-19 22:28:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-19 21:10:10.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-19 21:21:05.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-02-19 21:47:44.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-19 22:08:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-19 22:22:29.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-02-19 22:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-19 22:28:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"3f09dded8a1a72446beebfcafd715a6a3f83fb11","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2019-02-19 18:40:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e256903d9060761bae138805a4529b0946bea9b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-19 18:40:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3b33860aa43d7351d5d6f2a134426548c8e16e0","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2019-02-19 18:41:56.000000000","message":"Patch Set 1:\n\njust to make it clearer: i ran into this when my test failed and instead of nice failure message i got this trace:\n\n    Traceback (most recent call last):\n      File \"nova/tests/unit/virt/libvirt/test_config.py\", line 3716, in test_config_memory_balloon_driver_iommu\n        self.assertXmlEqual(expected_xml, xml)\n      File \"nova/tests/unit/virt/libvirt/test_config.py\", line 27, in assertXmlEqual\n        self.assertThat(actualXmlstr, matchers.XMLMatches(expectedXmlstr))\n      File \"/home/breton/src/openstack/nova/.tox/py27/lib/python2.7/site-packages/testtools/testcase.py\", line 496, in assertThat\n        mismatch_error \u003d self._matchHelper(matchee, matcher, message, verbose)\n      File \"/home/breton/src/openstack/nova/.tox/py27/lib/python2.7/site-packages/testtools/testcase.py\", line 547, in _matchHelper\n        mismatch \u003d matcher.match(matchee)\n      File \"nova/tests/unit/matchers.py\", line 427, in match\n        actual.getroot(), state, None)\n      File \"nova/tests/unit/matchers.py\", line 530, in _compare_node\n        actual_child_idx + 1)\n    UnboundLocalError: local variable \u0027actual_child_idx\u0027 referenced before assignment","accounts_in_message":[],"_revision_number":1},{"id":"4df55cf3857b074e927492c26c837354665199a8","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-02-19 19:18:57.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/637965/1/check/nova-out-of-tree-pvm/c5ad7ac : FAILURE in 37m 53s\n- nova-in-tree-pvm http://184.172.12.213/65/637965/1/check/nova-in-tree-pvm/6d136a9 : FAILURE in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"31a65062dea0d6dad2562a239441329be9a875d1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-19 19:59:41.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27896 : FAILURE in 1h 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5e812a1706f25c71f24e82fee7566435e6e298f4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-19 20:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/65/637965/1/check-vote/ext-nova-zuul/a9224f6 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"be4e787281f4e7f4d721cef53fc1cd9140fecdca","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-19 20:58:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/637965/1 : SUCCESS in 2h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fdfd94b53132554d96215a6e38d1a54d1bf26d18","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2019-02-19 21:09:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c12f2d77d5d07358712f2ba5d6f93da17855dd47","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-19 21:10:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c61c493b3acf1bab353519739912172102af690","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-19 21:21:05.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/637965/2 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3d027b58a53e76ac27a583f7b9c1d89ac082ad4a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-02-19 21:47:44.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/65/637965/2/check/nova-out-of-tree-pvm/0910b29 : FAILURE in 37m 21s\n- nova-in-tree-pvm http://184.172.12.213/65/637965/2/check/nova-in-tree-pvm/d84db22 : FAILURE in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f773faec6291b33653d8670066fd2ebdc96caf41","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-19 22:08:49.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/65/637965/2/check-vote/ext-nova-zuul/60a071a : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":2},{"id":"eb39efcf0224b0656e3647c2af7bc6120d213682","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-19 22:22:29.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-27899 : FAILURE in 1h 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4bbdcc625313527170be69ed02abad33e0fc72de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 22:26:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/637965/2/check/openstack-tox-cover/7a3f201/ : FAILURE in 13m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/637965/2/check/openstack-tox-lower-constraints/7df098f/ : FAILURE in 18m 16s\n- openstack-tox-pep8 http://logs.openstack.org/65/637965/2/check/openstack-tox-pep8/4031145/ : SUCCESS in 13m 22s\n- openstack-tox-py27 http://logs.openstack.org/65/637965/2/check/openstack-tox-py27/871c9b6/ : FAILURE in 18m 41s\n- openstack-tox-py35 http://logs.openstack.org/65/637965/2/check/openstack-tox-py35/a1d39f9/ : FAILURE in 18m 27s\n- openstack-tox-py36 http://logs.openstack.org/65/637965/2/check/openstack-tox-py36/7630a3e/ : FAILURE in 13m 31s\n- openstack-tox-docs http://logs.openstack.org/65/637965/2/check/openstack-tox-docs/5deb9e5/html/ : SUCCESS in 6m 51s\n- nova-tox-functional http://logs.openstack.org/65/637965/2/check/nova-tox-functional/cc893ae/ : SUCCESS in 18m 10s\n- nova-tox-functional-py35 http://logs.openstack.org/65/637965/2/check/nova-tox-functional-py35/04ebc31/ : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9fdc71eb56750b817864e45299ffeb82886ab38e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-19 22:28:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"53660f5455aee27915292ee9a1847b4db34de2b3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-19 22:50:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e908ba8306b81ed87aff52ebfb54630711d45d48","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2019-03-01 17:21:56.000000000","message":"Abandoned\n\nabandoned in favor of https://review.openstack.org/#/c/640411/1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08be016aec042558ca536fa3494cbc960b1cb734","revisions":{"6d0aebe32d0fc3a992459f19e9f75a629008e551":{"kind":"REWORK","_number":1,"created":"2019-02-19 18:40:23.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/65/637965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/637965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/637965/1"}}},"commit":{"parents":[{"commit":"d2a87a44468f91d5a468e1262e7e2fd8c7add35c","subject":"Merge \"create_veth_pair is unused, remove it.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2a87a44468f91d5a468e1262e7e2fd8c7add35c"}]}],"author":{"name":"Boris Bobrov","email":"breton@cynicmansion.ru","date":"2019-02-19 18:33:39.000000000","tz":60},"committer":{"name":"Boris Bobrov","email":"breton@cynicmansion.ru","date":"2019-02-19 18:40:21.000000000","tz":60},"subject":"Fallback to the first element if no nodes match","message":"Fallback to the first element if no nodes match\n\n\"UnboundLocalError: local variable \u0027actual_child_idx\u0027 referenced before\nassignment\" happens if the expected xml has a child element and actual\nhas not.\n\nChange-Id: I2773cfb5b3c2a13d514899c64dc052086134a5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d0aebe32d0fc3a992459f19e9f75a629008e551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d0aebe32d0fc3a992459f19e9f75a629008e551"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d2a87a44468f91d5a468e1262e7e2fd8c7add35c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"08be016aec042558ca536fa3494cbc960b1cb734":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-02-19 21:09:40.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/65/637965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/637965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/637965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/637965/2"}}},"commit":{"parents":[{"commit":"d2a87a44468f91d5a468e1262e7e2fd8c7add35c","subject":"Merge \"create_veth_pair is unused, remove it.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2a87a44468f91d5a468e1262e7e2fd8c7add35c"}]}],"author":{"name":"Boris Bobrov","email":"breton@cynicmansion.ru","date":"2019-02-19 18:33:39.000000000","tz":60},"committer":{"name":"Boris Bobrov","email":"breton@cynicmansion.ru","date":"2019-02-19 21:09:28.000000000","tz":0},"subject":"Don\u0027t reference unassigned variable in xml matchers in test","message":"Don\u0027t reference unassigned variable in xml matchers in test\n\n\"UnboundLocalError: local variable \u0027actual_child_idx\u0027 referenced before\nassignment\" happens if the expected xml has a child element and actual\nhas not.\n\nFallback to the first element if no nodes match.\n\nChange-Id: I2773cfb5b3c2a13d514899c64dc052086134a5c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08be016aec042558ca536fa3494cbc960b1cb734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08be016aec042558ca536fa3494cbc960b1cb734"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d2a87a44468f91d5a468e1262e7e2fd8c7add35c","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
