)]}'
{"id":"openstack%2Fnova~638803","triplet_id":"openstack%2Fnova~stable%2Fqueens~I44f09aec60b0b18c458f9ba6d8b725db962e9cc7","project":"openstack/nova","branch":"stable/queens","topic":"bug/1805164","hashtags":[],"change_id":"I44f09aec60b0b18c458f9ba6d8b725db962e9cc7","subject":"Provide a useful error message when trying to update non-compute services","status":"MERGED","created":"2019-02-23 10:13:45.000000000","updated":"2019-03-15 15:49:53.000000000","submitted":"2019-03-15 15:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"638803-1552664993015-73b52645","meta_rev_id":"8a119299aa03898ab7d35cf686b0f53ee5666452","_number":638803,"virtual_id_number":638803,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2019-02-24 03:44:54.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-15 11:53:20.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-02-23 11:37:02.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-15 12:38:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-03-15 15:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-15 11:08:15.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2019-03-12 20:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-12 19:13:42.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2019-03-12 20:28:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-23 11:37:02.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-24 03:44:54.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-12 19:13:42.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-03-12 20:28:09.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-03-15 11:53:20.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-03-15 12:38:28.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-15 15:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd95571d64f59a93969b1e123e0be307a342ef6e","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-02-23 10:13:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"cae4d29f7188a45f80ee3def35ec1c8ffefddd38","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-23 10:57:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/638803/1 : FAILURE in 42m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3567017da3ce7832a8a3962ee4239a3b44413586","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-23 11:15:37.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial/74a9674/ : SUCCESS in 1h 01m 09s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial-py3/84446ec/ : FAILURE in 56m 39s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/grenade-dsvm-xenial/dfe3370/ : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"161f9b3f80f728275203a990887c8f0fde48a9e7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-02-23 11:37:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/03/638803/1/check-nova/Nova-ML2-Sriov/b2d1550 : FAILURE in 31m 12s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/03/638803/1/check-nova/Nova-MACVTAP-ML2-Sriov/225d99b : FAILURE in 28m 13s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":1},{"id":"6b1dadf4ead124114a5ff971981d8dbbaeb7e9cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-23 11:52:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/check/openstack-tox-pep8/c66f492/ : SUCCESS in 8m 22s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/check/openstack-tox-py27/efbaa2f/ : SUCCESS in 14m 15s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/check/openstack-tox-py35/7d58976/ : SUCCESS in 12m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/check/build-openstack-sphinx-docs/77f6547/html/ : SUCCESS in 5m 38s\n- tempest-full http://logs.openstack.org/03/638803/1/check/tempest-full/ea804b9/ : SUCCESS in 1h 19m 56s\n- neutron-grenade http://logs.openstack.org/03/638803/1/check/neutron-grenade/913c5f9/ : SUCCESS in 1h 03m 50s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/check/tempest-full-py3/4085ec9/ : SUCCESS in 1h 06m 37s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/03/638803/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6d72cef/ : SUCCESS in 45m 01s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/check/nova-cells-v1/bf42c1e/ : SUCCESS in 43m 15s\n- nova-live-migration http://logs.openstack.org/03/638803/1/check/nova-live-migration/3dffb8c/ : SUCCESS in 44m 26s\n- nova-multiattach http://logs.openstack.org/03/638803/1/check/nova-multiattach/918b260/ : SUCCESS in 51m 51s\n- nova-next http://logs.openstack.org/03/638803/1/check/nova-next/d857e35/ : SUCCESS in 1h 23m 53s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/check/nova-tox-functional/e2d47c4/ : SUCCESS in 17m 19s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/check/nova-tox-functional-py35/26f8d88/ : SUCCESS in 20m 22s\n- neutron-grenade-multinode http://logs.openstack.org/03/638803/1/check/neutron-grenade-multinode/50d1991/ : SUCCESS in 56m 21s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/03/638803/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/5b9e0ea/ : FAILURE in 57m 05s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/03/638803/1/check/devstack-plugin-ceph-tempest/db6bd50/ : SUCCESS in 1h 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5edff6909adab547cb4e3ee27e20243bf01f9985","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-24 03:44:54.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/638803/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2eff7ef : FAILURE in 2m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b4cf78a65835295beb807b8f64479697591d58f1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-12 19:13:01.000000000","message":"Topic changed from bug/1805164-stable/rocky-stable/queens to bug/1805164","accounts_in_message":[],"_revision_number":1},{"id":"072bbbc7c8f713fc0fc8bc5920af707e270a1d44","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-12 19:13:42.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport, lgtm.","accounts_in_message":[],"_revision_number":1},{"id":"ee482e5b5d37fd886bd6d2f4c214a07265f51833","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-03-12 20:28:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"904019973a56e03353216dc457873e57fd6918ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 21:39:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9b42a4748555ea9ef351aa3adec834658b9739c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 04:08:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/gate/openstack-tox-pep8/fb44544/ : SUCCESS in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py27/ce37597/ : SUCCESS in 13m 48s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py35/85c0714/ : SUCCESS in 14m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/gate/build-openstack-sphinx-docs/f7dd798/html/ : SUCCESS in 9m 10s\n- tempest-full http://logs.openstack.org/03/638803/1/gate/tempest-full/254f0c0/ : SUCCESS in 1h 21m 44s\n- neutron-grenade http://logs.openstack.org/03/638803/1/gate/neutron-grenade/e1e9e49/ : SUCCESS in 1h 19m 49s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/gate/tempest-full-py3/05cb2ee/ : SUCCESS in 1h 19m 50s\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/gate/nova-cells-v1/c01ee5c/ : SUCCESS in 53m 10s\n- nova-live-migration http://logs.openstack.org/03/638803/1/gate/nova-live-migration/597e48f/ : SUCCESS in 45m 01s\n- nova-multiattach http://logs.openstack.org/03/638803/1/gate/nova-multiattach/a0ea85a/ : SUCCESS in 57m 08s\n- nova-next http://logs.openstack.org/03/638803/1/gate/nova-next/a489179/ : SUCCESS in 1h 23m 05s\n- tempest-slow http://logs.openstack.org/03/638803/1/gate/tempest-slow/2d95d1d/ : FAILURE in 1h 53m 41s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/gate/nova-tox-functional/96e9a79/ : SUCCESS in 20m 47s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/gate/nova-tox-functional-py35/952a2bb/ : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"010007dbad266a184b735d146d252103427f6b8f","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-03-13 06:40:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15bf4ec70c0d42edc7b752d60742bad8ce00c2a9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-13 07:38:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/638803/1 : FAILURE in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"140f0e3a1158044f86619788e72b5b00f92d0946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 08:29:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/check/openstack-tox-pep8/c171037/ : SUCCESS in 8m 55s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/check/openstack-tox-py27/c45a9bd/ : SUCCESS in 12m 57s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/check/openstack-tox-py35/9f35fbd/ : SUCCESS in 12m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/check/build-openstack-sphinx-docs/3d67a7e/html/ : SUCCESS in 5m 37s\n- tempest-full http://logs.openstack.org/03/638803/1/check/tempest-full/2636f5e/ : SUCCESS in 1h 29m 46s\n- neutron-grenade http://logs.openstack.org/03/638803/1/check/neutron-grenade/106542a/ : SUCCESS in 52m 13s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/check/tempest-full-py3/d8bf227/ : SUCCESS in 1h 32m 47s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/03/638803/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ae70cb1/ : SUCCESS in 52m 45s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/check/nova-cells-v1/76ab59f/ : SUCCESS in 45m 13s\n- nova-live-migration http://logs.openstack.org/03/638803/1/check/nova-live-migration/1f12209/ : SUCCESS in 44m 26s\n- nova-multiattach http://logs.openstack.org/03/638803/1/check/nova-multiattach/41875d4/ : SUCCESS in 55m 02s\n- nova-next http://logs.openstack.org/03/638803/1/check/nova-next/6265d66/ : SUCCESS in 1h 25m 20s\n- tempest-slow http://logs.openstack.org/03/638803/1/check/tempest-slow/5f32032/ : FAILURE in 1h 42m 58s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/check/nova-tox-functional/34c1dd6/ : SUCCESS in 30m 59s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/check/nova-tox-functional-py35/599b627/ : SUCCESS in 18m 02s\n- neutron-grenade-multinode http://logs.openstack.org/03/638803/1/check/neutron-grenade-multinode/c7290ce/ : SUCCESS in 1h 02m 58s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/03/638803/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/24edd26/ : FAILURE in 1h 28m 25s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/03/638803/1/check/devstack-plugin-ceph-tempest/a914539/ : FAILURE in 1h 11m 40s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/03/638803/1/check/tempest-multinode-full/b394652/ : SUCCESS in 1h 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67c212711c4fd8b51a311c5ac66e097c27ce4ed4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-13 10:52:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial/eb407e8/ : SUCCESS in 58m 54s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial-py3/62b9e15/ : FAILURE in 55m 01s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/grenade-dsvm-xenial/3d3aa6b/ : SUCCESS in 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"529c2d646ad0f957de8ad45d5b98b6c1016158d5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-13 14:53:37.000000000","message":"Patch Set 1:\n\nrecheck bug 1819910","accounts_in_message":[],"_revision_number":1},{"id":"2a0518a7b3d34d011cc023f49f7ffc7637b435f6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-13 15:41:45.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/638803/1 : FAILURE in 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7943a7ea748c88882e3683318cbebc7346017e73","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-13 17:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial/99418c3/ : SUCCESS in 1h 02m 46s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial-py3/72a4fae/ : FAILURE in 1h 07m 25s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/grenade-dsvm-xenial/68683f4/ : SUCCESS in 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8e66095ac340fe81bc43e99a2c7becf78745eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 18:41:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/check/openstack-tox-pep8/3668ee8/ : SUCCESS in 10m 33s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/check/openstack-tox-py27/46c230a/ : SUCCESS in 13m 11s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/check/openstack-tox-py35/0b8090d/ : SUCCESS in 14m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/check/build-openstack-sphinx-docs/16a5582/html/ : SUCCESS in 7m 23s\n- tempest-full http://logs.openstack.org/03/638803/1/check/tempest-full/9dea603/ : SUCCESS in 1h 28m 08s\n- neutron-grenade http://logs.openstack.org/03/638803/1/check/neutron-grenade/401da01/ : SUCCESS in 52m 52s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/check/tempest-full-py3/5356a3d/ : SUCCESS in 1h 18m 37s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/03/638803/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5f152ce/ : SUCCESS in 50m 56s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/check/nova-cells-v1/040b693/ : SUCCESS in 43m 22s\n- nova-live-migration http://logs.openstack.org/03/638803/1/check/nova-live-migration/42d74c3/ : SUCCESS in 47m 54s\n- nova-multiattach http://logs.openstack.org/03/638803/1/check/nova-multiattach/f8a0ff7/ : SUCCESS in 1h 04m 25s\n- nova-next http://logs.openstack.org/03/638803/1/check/nova-next/8434c24/ : SUCCESS in 1h 27m 56s\n- tempest-slow http://logs.openstack.org/03/638803/1/check/tempest-slow/adf3cd2/ : SUCCESS in 2h 03m 00s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/check/nova-tox-functional/f792de5/ : SUCCESS in 22m 11s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/check/nova-tox-functional-py35/55f9855/ : SUCCESS in 25m 27s\n- neutron-grenade-multinode http://logs.openstack.org/03/638803/1/check/neutron-grenade-multinode/c176852/ : SUCCESS in 1h 07m 19s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/03/638803/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/492b6d2/ : FAILURE in 1h 02m 44s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/03/638803/1/check/devstack-plugin-ceph-tempest/d8738c9/ : FAILURE in 1h 15m 20s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/03/638803/1/check/tempest-multinode-full/d681ebe/ : SUCCESS in 1h 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f11faa1eccfd9b53543ab4e4616981d9e76fbcb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 18:41:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85ab3923bbdbe4a79a88e570f93f50980c7f946e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 21:03:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/gate/openstack-tox-pep8/17dcbc1/ : SUCCESS in 9m 39s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py27/3690d35/ : SUCCESS in 15m 00s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py35/d322373/ : SUCCESS in 14m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/gate/build-openstack-sphinx-docs/fc2ae12/html/ : SUCCESS in 11m 16s\n- tempest-full http://logs.openstack.org/03/638803/1/gate/tempest-full/0b60944/ : SUCCESS in 1h 24m 16s\n- neutron-grenade http://logs.openstack.org/03/638803/1/gate/neutron-grenade/c9a9f86/ : SUCCESS in 56m 55s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/gate/tempest-full-py3/119b263/ : SUCCESS in 1h 13m 51s\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/gate/nova-cells-v1/ed8c7c2/ : SUCCESS in 51m 58s\n- nova-live-migration http://logs.openstack.org/03/638803/1/gate/nova-live-migration/090a46a/ : SUCCESS in 46m 38s\n- nova-multiattach http://logs.openstack.org/03/638803/1/gate/nova-multiattach/2625e21/ : SUCCESS in 1h 00m 43s\n- nova-next http://logs.openstack.org/03/638803/1/gate/nova-next/91ae4a5/ : SUCCESS in 1h 22m 53s\n- tempest-slow http://logs.openstack.org/03/638803/1/gate/tempest-slow/3c9875a/ : FAILURE in 1h 45m 09s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/gate/nova-tox-functional/722b5ee/ : SUCCESS in 18m 25s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/gate/nova-tox-functional-py35/99c44a7/ : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"32a933cd9fc003f81b8de587d867a9d0a2e6df94","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-03-15 11:08:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5937676a97f95c5b99c53439e7f7d4fcc478d35f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-15 11:53:20.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/638803/1 : FAILURE in 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aa5ff5d94d4c269e075f2ed19a4bbc9a40b3c617","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-15 12:38:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial/80ed9c0/ : SUCCESS in 1h 03m 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/tempest-dsvm-full-xenial-py3/8617e61/ : FAILURE in 1h 00m 14s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/03/638803/1/check/grenade-dsvm-xenial/7f9c235/ : SUCCESS in 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f66520c868ea6621979fd352b71044fb3ed7f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 14:06:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/check/openstack-tox-pep8/a6f6435/ : SUCCESS in 10m 50s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/check/openstack-tox-py27/6514aab/ : SUCCESS in 14m 51s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/check/openstack-tox-py35/70cee27/ : SUCCESS in 16m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/check/build-openstack-sphinx-docs/69afd3a/html/ : SUCCESS in 7m 05s\n- tempest-full http://logs.openstack.org/03/638803/1/check/tempest-full/764ac4a/ : SUCCESS in 1h 21m 26s\n- neutron-grenade http://logs.openstack.org/03/638803/1/check/neutron-grenade/c4aa14f/ : SUCCESS in 52m 42s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/check/tempest-full-py3/46c35a0/ : SUCCESS in 1h 19m 59s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/03/638803/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ed2053d/ : SUCCESS in 47m 01s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/check/nova-cells-v1/1fc0436/ : SUCCESS in 45m 28s\n- nova-live-migration http://logs.openstack.org/03/638803/1/check/nova-live-migration/b7e57e0/ : SUCCESS in 47m 36s\n- nova-multiattach http://logs.openstack.org/03/638803/1/check/nova-multiattach/9ebfe1e/ : SUCCESS in 1h 01m 20s\n- nova-next http://logs.openstack.org/03/638803/1/check/nova-next/cef8739/ : SUCCESS in 1h 34m 00s\n- tempest-slow http://logs.openstack.org/03/638803/1/check/tempest-slow/93cb885/ : SUCCESS in 1h 53m 10s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/check/nova-tox-functional/3492a7f/ : SUCCESS in 24m 32s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/check/nova-tox-functional-py35/d936d2a/ : SUCCESS in 20m 53s\n- neutron-grenade-multinode http://logs.openstack.org/03/638803/1/check/neutron-grenade-multinode/a5fddea/ : SUCCESS in 58m 28s\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/03/638803/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/d428610/ : FAILURE in 1h 10m 00s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/03/638803/1/check/devstack-plugin-ceph-tempest/2842d6c/ : FAILURE in 1h 17m 48s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/03/638803/1/check/tempest-multinode-full/a6bd207/ : SUCCESS in 1h 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdac21cfc87878f2c7fbbe4ef5767c6e00ede16c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 14:06:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfe871c95e49437b511d99ef253d536c3090c666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 15:49:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/638803/1/gate/openstack-tox-pep8/e4ae811/ : SUCCESS in 10m 11s\n- openstack-tox-py27 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py27/7faf1de/ : SUCCESS in 14m 01s\n- openstack-tox-py35 http://logs.openstack.org/03/638803/1/gate/openstack-tox-py35/2eafbbf/ : SUCCESS in 14m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/638803/1/gate/build-openstack-sphinx-docs/4cfc1fc/html/ : SUCCESS in 8m 03s\n- tempest-full http://logs.openstack.org/03/638803/1/gate/tempest-full/cad27fa/ : SUCCESS in 1h 29m 04s\n- neutron-grenade http://logs.openstack.org/03/638803/1/gate/neutron-grenade/6575f74/ : SUCCESS in 55m 14s\n- tempest-full-py3 http://logs.openstack.org/03/638803/1/gate/tempest-full-py3/6120e5f/ : SUCCESS in 1h 30m 20s\n- nova-cells-v1 http://logs.openstack.org/03/638803/1/gate/nova-cells-v1/18270a4/ : SUCCESS in 57m 06s\n- nova-live-migration http://logs.openstack.org/03/638803/1/gate/nova-live-migration/40adf05/ : SUCCESS in 58m 16s\n- nova-multiattach http://logs.openstack.org/03/638803/1/gate/nova-multiattach/ecb9419/ : SUCCESS in 55m 24s\n- nova-next http://logs.openstack.org/03/638803/1/gate/nova-next/3a27e9d/ : SUCCESS in 1h 13m 01s\n- tempest-slow http://logs.openstack.org/03/638803/1/gate/tempest-slow/0da60a8/ : SUCCESS in 1h 40m 54s\n- nova-tox-functional http://logs.openstack.org/03/638803/1/gate/nova-tox-functional/fd80c4d/ : SUCCESS in 20m 26s\n- nova-tox-functional-py35 http://logs.openstack.org/03/638803/1/gate/nova-tox-functional-py35/b2c1e9e/ : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8a119299aa03898ab7d35cf686b0f53ee5666452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 15:49:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b041160c8d56f657059743bc5a0e6dbefc27ddd6","revisions":{"b041160c8d56f657059743bc5a0e6dbefc27ddd6":{"kind":"REWORK","_number":1,"created":"2019-02-23 10:13:45.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/03/638803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/03/638803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/03/638803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/03/638803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/03/638803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/03/638803/1"}}},"commit":{"parents":[{"commit":"09456e656aec47d18d48c05fb33d7fb044bd076e","subject":"Merge \"PCI: do not force remove allocated devices\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09456e656aec47d18d48c05fb33d7fb044bd076e"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-11-28 17:52:17.000000000","tz":-300},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-02-23 10:13:45.000000000","tz":0},"subject":"Provide a useful error message when trying to update non-compute services","message":"Provide a useful error message when trying to update non-compute services\n\nStarting in Pike, we disallowed trying to update (enable/disable/force down)\nnon-nova-compute services because of multi-cell support using host mappings\nto lookup service records, and simply because disabling non-compute services\ndoesn\u0027t do anything.\n\nHowever, before microversion 2.53, the error the user gets back is confusing:\n\n  HTTP exception thrown: Host \u0027p024.domain.com\u0027 is not mapped to any cell\n\nThis change provides a useful error message in this case and also changes\nthe 404 response to a 400 response to align with the type of error and the\nbehavior of the 2.53 microversion.\n\nChange-Id: I44f09aec60b0b18c458f9ba6d8b725db962e9cc7\nCloses-Bug: #1805164\n(cherry picked from commit 95cbc35b0bdae0de15569bc35fadbcf970da5b9e)\n(cherry picked from commit 315d9fab7aaa6b9835e80f5f56b49b44d382ca1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b041160c8d56f657059743bc5a0e6dbefc27ddd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b041160c8d56f657059743bc5a0e6dbefc27ddd6"}]},"parents_data":[{"branch_name":"refs/heads/stable/queens","commit_id":"09456e656aec47d18d48c05fb33d7fb044bd076e","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
