)]}'
{"id":"openstack%2Fnova~639091","triplet_id":"openstack%2Fnova~master~If2035cac931c42c440d61ba97ebc7e9e92141a28","project":"openstack/nova","branch":"master","topic":"bug/1353939","hashtags":[],"change_id":"If2035cac931c42c440d61ba97ebc7e9e92141a28","subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","status":"MERGED","created":"2019-02-25 13:39:08.000000000","updated":"2019-06-25 15:51:09.000000000","submitted":"2019-05-31 15:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639091-1559317286041-61dee02a","meta_rev_id":"0d12f0635805d06d9820c2bcf5f452ddfaff5d9e","_number":639091,"virtual_id_number":639091,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-29 16:00:41.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-29 15:50:36.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"date":"2019-05-30 03:27:24.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-29 20:44:37.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-30 01:38:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-29 17:09:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-05-31 15:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-29 21:52:22.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"date":"2019-05-29 18:03:50.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2019-05-29 15:38:09.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-30 06:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-31 12:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":1,"date":"2019-05-31 12:35:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-05-29 21:47:54.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-05-30 08:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-05-29 15:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-31 12:54:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-25 13:40:10.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-02-26 04:45:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 12:14:19.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-01 14:08:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 15:38:09.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 15:39:34.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2019-05-29 15:50:36.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 16:00:41.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 17:09:36.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-05-29 18:03:50.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-29 20:44:37.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 21:47:54.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-29 21:52:22.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 01:38:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 03:27:24.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 06:14:45.000000000","updated_by":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"reviewer":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"state":"REVIEWER"},{"updated":"2019-05-30 08:35:13.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-05-31 12:35:29.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2019-05-31 12:54:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-31 15:41:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"669a21e29d520142953022b6b874e5227743c048","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-02-25 13:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9fdfe6ce5d8bf5f3a360953376caad9cc74f111","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-25 13:39:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"625828c09f24dca62db41dd9f7efa2eff24060c1","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-02-25 13:41:42.000000000","message":"Patch Set 1:\n\nNeed to fix the following unit tests:\n\n  - test_private_destroy_ebusy_timeout\n  - test_private_destroy_ebusy_multiple_attempt_ok","accounts_in_message":[],"_revision_number":1},{"id":"69b13fda37facbf06f820f25d87feee450ab3705","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-02-25 13:42:08.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ea5c9815d2a5391fc28d7be6cd1a20d6f1a8ac4c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-25 14:14:24.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28212 : FAILURE in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dcd2ebce345b36be6dc3c5a017065d7df1696c62","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-25 14:50:44.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/1/check-vote/ext-nova-zuul/93ee8a6 : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8aa50d7217ba05e3ddda377407a7da96b4d0be38","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-02-25 15:10:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfbc7d03461a84aa865c028ad25d6e05a41206a3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-25 15:10:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64c67528b89830cc7fa3bafbd1fbb141b92aec40","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-25 15:41:58.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28217 : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6786fb93586658f97231e7eeb6b69490d8fd137f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-25 16:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/2/check-vote/ext-nova-zuul/fd2b94e : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":2},{"id":"fc90ade07dfc7c231ddb92edccbda0fc69214259","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-02-25 16:18:11.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"eea7889c38445c138bb50aa11c1766bacc3fbc99","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-25 16:34:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/2/check/tempest-dsvm-full-xenial/ab775dc/ : SUCCESS in 1h 07m 56s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/2/check/tempest-dsvm-full-xenial-py3/35e8407/ : SUCCESS in 1h 23m 09s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/2/check/grenade-dsvm-xenial/02f4bee/ : SUCCESS in 1h 01m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"490206de83e6ed0154f52da9ea1839fff516d354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-25 17:38:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/639091/2/check/grenade-py3/0bf15e7/ : SUCCESS in 1h 12m 34s\n- tempest-full-py3 http://logs.openstack.org/91/639091/2/check/tempest-full-py3/a893369/ : SUCCESS in 1h 22m 30s\n- openstack-tox-cover http://logs.openstack.org/91/639091/2/check/openstack-tox-cover/5ffebd6/ : FAILURE in 16m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/2/check/openstack-tox-lower-constraints/cd3a758/ : FAILURE in 17m 20s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/2/check/openstack-tox-pep8/a4ab738/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/2/check/openstack-tox-py27/606b7c8/ : FAILURE in 13m 10s\n- openstack-tox-py35 http://logs.openstack.org/91/639091/2/check/openstack-tox-py35/70dc217/ : FAILURE in 14m 57s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/2/check/openstack-tox-py36/fa8ab0b/ : FAILURE in 16m 54s\n- openstack-tox-docs http://logs.openstack.org/91/639091/2/check/openstack-tox-docs/5156ab0/html/ : SUCCESS in 6m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2c5231f/ : SUCCESS in 53m 26s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/2/check/devstack-plugin-ceph-tempest/87bb6f7/ : TIMED_OUT in 2h 04m 51s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/91/639091/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/5078f47/ : SUCCESS in 1h 08m 46s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/2/check/neutron-grenade-multinode/ce9b05d/ : SUCCESS in 1h 19m 34s\n- nova-live-migration http://logs.openstack.org/91/639091/2/check/nova-live-migration/aa96053/ : SUCCESS in 50m 08s\n- nova-lvm http://logs.openstack.org/91/639091/2/check/nova-lvm/348a7a4/ : SUCCESS in 1h 00m 54s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/2/check/nova-next/f52e6ec/ : SUCCESS in 2h 06m 07s\n- nova-tox-functional http://logs.openstack.org/91/639091/2/check/nova-tox-functional/960e502/ : SUCCESS in 22m 43s\n- nova-tox-functional-py35 http://logs.openstack.org/91/639091/2/check/nova-tox-functional-py35/56bc3f8/ : SUCCESS in 21m 51s\n- tempest-multinode-full http://logs.openstack.org/91/639091/2/check/tempest-multinode-full/fc80461/ : SUCCESS in 2h 13m 55s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/91/639091/2/check/tempest-slow-py3/d5bdede/ : SUCCESS in 2h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9547713a5cd7112b3bb5478d8651718c905e8301","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-25 18:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/2 : SUCCESS in 2h 52m 28s","accounts_in_message":[],"_revision_number":2},{"id":"72778bd269a83dac0a8c743a0bb066ecf4922b2e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-25 18:08:01.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/efcc50c : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a354c4688e6f4a5870c10c86763c6be6ebd025dc","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-02-26 04:45:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/91/639091/2/check-nova/Nova-ML2-Sriov/ff4ba80 : SUCCESS in 1h 28m 16s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/91/639091/2/check-nova/Nova-MACVTAP-ML2-Sriov/8f5147c : SUCCESS in 55m 26s (non-voting)\n- NVMe http://13.74.249.42/91/639091/2/check-nova/NVMe/096d7c6 : SUCCESS in 45m 34s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"3b357fdd0ee44bddb4c1ccd539408abda026a5c7","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-08 13:30:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d48b1a435b97311b0c398a4208d1504c91c12c7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 14:30:10.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28965 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1ebe2f43d02f7cdadad31934cfb564ea1851e60c","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-03-08 15:14:31.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM although not my area of expertise.\n\nThis does make _destroy() an even longer method than before though, so if possible it would be nice to extract out the libvirt.VIR_ERR_SYSTEM_ERROR clause into a separate method.  This would also reduce the indentation providing more width and increasing readability.","accounts_in_message":[],"_revision_number":3},{"id":"2b5f085aa10eaa5e87b5d9e1fdd32b5546b58a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 15:51:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/639091/3/check/grenade-py3/7720ff2/ : SUCCESS in 57m 17s\n- tempest-full-py3 http://logs.openstack.org/91/639091/3/check/tempest-full-py3/3090149/ : SUCCESS in 1h 19m 06s\n- openstack-tox-cover http://logs.openstack.org/91/639091/3/check/openstack-tox-cover/9c8a647/cover/ : SUCCESS in 18m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/3/check/openstack-tox-lower-constraints/b6534f9/ : SUCCESS in 12m 25s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/3/check/openstack-tox-pep8/076a576/ : SUCCESS in 11m 20s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/3/check/openstack-tox-py27/f320637/ : SUCCESS in 12m 03s\n- openstack-tox-py35 http://logs.openstack.org/91/639091/3/check/openstack-tox-py35/32db53e/ : SUCCESS in 13m 15s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/3/check/openstack-tox-py36/8b9c1f0/ : SUCCESS in 11m 19s\n- openstack-tox-docs http://logs.openstack.org/91/639091/3/check/openstack-tox-docs/296bee0/html/ : SUCCESS in 6m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/50a2f94/ : SUCCESS in 43m 10s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/3/check/devstack-plugin-ceph-tempest/ad8efe4/ : SUCCESS in 1h 22m 13s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/3/check/neutron-grenade-multinode/45265cb/ : SUCCESS in 1h 13m 27s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/3/check/nova-grenade-live-migration/278260d/ : SUCCESS in 1h 06m 52s\n- nova-live-migration http://logs.openstack.org/91/639091/3/check/nova-live-migration/f3f8486/ : FAILURE in 44m 10s\n- nova-lvm http://logs.openstack.org/91/639091/3/check/nova-lvm/0a5eb3f/ : SUCCESS in 1h 00m 14s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/3/check/nova-next/c2ae0f8/ : SUCCESS in 1h 43m 03s\n- nova-tox-functional http://logs.openstack.org/91/639091/3/check/nova-tox-functional/bb4462e/ : SUCCESS in 20m 38s\n- nova-tox-functional-py35 http://logs.openstack.org/91/639091/3/check/nova-tox-functional-py35/bbd75b4/ : SUCCESS in 16m 52s\n- tempest-multinode-full http://logs.openstack.org/91/639091/3/check/tempest-multinode-full/a2180f0/ : SUCCESS in 1h 52m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/91/639091/3/check/tempest-slow-py3/f0d3bf0/ : SUCCESS in 2h 14m 06s","accounts_in_message":[],"_revision_number":3},{"id":"d6e1b297cf15066092151354514c850a5141562f","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-08 18:24:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef5a56b71f1ebf644a3e675cf4594e4f5ad59237","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 19:17:45.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28977 : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":3},{"id":"876c8515f54a73d463b663fa24381d5e52e94ca2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-08 19:35:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/tempest-dsvm-full-xenial/e31258f/ : SUCCESS in 1h 05m 34s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/tempest-dsvm-full-xenial-py3/29ee330/ : SUCCESS in 1h 10m 59s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/grenade-dsvm-xenial/55e49d0/ : SUCCESS in 55m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"281a8cd2eaa038ad1ff7ef29efda27ac2338f143","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 19:59:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/3 : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":3},{"id":"317809f903f335000ca208a69a7c07b88c1bd923","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 22:49:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/639091/3/check/grenade-py3/0d491fd/ : SUCCESS in 1h 01m 10s\n- tempest-full-py3 http://logs.openstack.org/91/639091/3/check/tempest-full-py3/c0eb194/ : SUCCESS in 1h 32m 48s\n- openstack-tox-cover http://logs.openstack.org/91/639091/3/check/openstack-tox-cover/b850a1a/cover/ : SUCCESS in 17m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/3/check/openstack-tox-lower-constraints/6a3d287/ : SUCCESS in 13m 38s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/3/check/openstack-tox-pep8/8c35be0/ : SUCCESS in 10m 16s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/3/check/openstack-tox-py27/1184efc/ : SUCCESS in 12m 59s\n- openstack-tox-py35 http://logs.openstack.org/91/639091/3/check/openstack-tox-py35/e849583/ : SUCCESS in 16m 23s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/3/check/openstack-tox-py36/0f88c5b/ : SUCCESS in 12m 15s\n- openstack-tox-docs http://logs.openstack.org/91/639091/3/check/openstack-tox-docs/91276ba/html/ : SUCCESS in 6m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7d1cb4e/ : SUCCESS in 55m 55s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/3/check/devstack-plugin-ceph-tempest/f4498b0/ : SUCCESS in 1h 24m 53s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/3/check/neutron-grenade-multinode/84b6360/ : SUCCESS in 1h 06m 36s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/3/check/nova-grenade-live-migration/be38232/ : SUCCESS in 1h 05m 27s\n- nova-live-migration http://logs.openstack.org/91/639091/3/check/nova-live-migration/c3fdd8c/ : FAILURE in 50m 05s\n- nova-lvm http://logs.openstack.org/91/639091/3/check/nova-lvm/296caff/ : SUCCESS in 1h 08m 36s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/3/check/nova-next/4922e9a/ : SUCCESS in 2h 30m 39s\n- nova-tox-functional http://logs.openstack.org/91/639091/3/check/nova-tox-functional/93c25b4/ : SUCCESS in 21m 24s\n- nova-tox-functional-py35 http://logs.openstack.org/91/639091/3/check/nova-tox-functional-py35/19b4e26/ : SUCCESS in 20m 53s\n- tempest-multinode-full http://logs.openstack.org/91/639091/3/check/tempest-multinode-full/e35cd60/ : SUCCESS in 1h 55m 12s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/91/639091/3/check/tempest-slow-py3/4aa42bc/ : SUCCESS in 1h 56m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b300a0936c79362f43fd36dcca016f9d3d58c023","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-09 10:10:59.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/3/check-vote/ext-nova-zuul/dafeafe : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b22fd57fb42dedfc3eeb28fc8493eb1d603c0d2f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 12:29:57.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/cd59b7a : FAILURE in 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"669426743bcbb17a65d0f842cf27fef16b415a97","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-19 10:46:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f78c0302212933beb50ebb8c9d91362901759869","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-19 10:47:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23c6280c555f5fd557f0a915a64322ee2aef4e68","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-19 11:27:15.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/3/check/nova-out-of-tree-pvm/10e5b76 : FAILURE in 39m 40s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/3/check/nova-in-tree-pvm/82a4186 : FAILURE in 38m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7875d603e0da9b4924b3da9d43c6991c5d1992aa","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-19 11:41:44.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29348 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2b67edef0f137bb7ac774d8c1173d173fee5e5d3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-19 12:01:01.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/tempest-dsvm-full-xenial/3e46b65/ : SUCCESS in 1h 13m 13s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/tempest-dsvm-full-xenial-py3/206d98c/ : SUCCESS in 1h 13m 21s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/3/check/grenade-dsvm-xenial/64f71b2/ : SUCCESS in 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3169402c9721a9176fa7a85f844880299e4ead0f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-19 12:11:03.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/3/check-vote/ext-nova-zuul/d221b34 : SUCCESS in 1h 23m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e42021e9385128b3f8bc89a8917977630776b6d6","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-19 12:49:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b1b6545311ba15ee230ccd2f5618bc478c92310d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-19 12:50:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7f9c2ebe09ea8434ffa747f03023c44346b03b4c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-19 13:01:02.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29355 : FAILURE in 11m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2a8410a53df080b202ed4d3ccf402ad598760c61","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-03-19 13:27:23.000000000","message":"Patch Set 4:\n\n\u003e LGTM although not my area of expertise.\n \u003e \n \u003e This does make _destroy() an even longer method than before though,\n \u003e so if possible it would be nice to extract out the\n \u003e libvirt.VIR_ERR_SYSTEM_ERROR clause into a separate method.  This\n \u003e would also reduce the indentation providing more width and\n \u003e increasing readability.\n\nI see where you\u0027re coming from, but two thoughts to consider:\n\n- The EBUSY error is handled only in one place\n- Splitting it out into a separate method, while slightly improving readability, introduces yet another indirection\n\nSo it\u0027s a trade-off: readability vs. indirection.  For now, I\u0027ll lean towards a \"slightly longer method\" at the cost of indirection.","accounts_in_message":[],"_revision_number":4},{"id":"349058c77ca05b9295c558bdbff7774ed7120b6d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-19 13:28:57.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/4/check/nova-out-of-tree-pvm/425ec34 : FAILURE in 38m 02s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/4/check/nova-in-tree-pvm/9714944 : FAILURE in 38m 27s","accounts_in_message":[],"_revision_number":4},{"id":"c40b07030d232a1214f132c464b505b45af21dd4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-19 14:10:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/4/check/tempest-dsvm-full-xenial/40d334d/ : SUCCESS in 1h 16m 02s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/4/check/tempest-dsvm-full-xenial-py3/c154834/ : SUCCESS in 1h 20m 25s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/4/check/grenade-dsvm-xenial/51da0f0/ : SUCCESS in 56m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d539326ad185c02b96d1e3cbce57a237acfb400","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-19 14:22:56.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/4/check-vote/ext-nova-zuul/92fd9ac : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9887d1459bd6c7848779895adfb8e275b70adde6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-19 15:05:03.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639091/4 : FAILURE in 2h 14m 38s","accounts_in_message":[],"_revision_number":4},{"id":"ef511eb2bcef5835b04f1d90a3fa82557628fd20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 15:16:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/639091/4/check/grenade-py3/88bee8a/ : SUCCESS in 58m 07s\n- tempest-full-py3 http://logs.openstack.org/91/639091/4/check/tempest-full-py3/b835653/ : SUCCESS in 1h 31m 06s\n- openstack-tox-cover http://logs.openstack.org/91/639091/4/check/openstack-tox-cover/85af5fa/cover/ : SUCCESS in 15m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/4/check/openstack-tox-lower-constraints/f5eb89c/ : SUCCESS in 12m 41s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/4/check/openstack-tox-pep8/b0c8a78/ : SUCCESS in 10m 47s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/4/check/openstack-tox-py27/f16a089/ : SUCCESS in 11m 59s\n- openstack-tox-py35 http://logs.openstack.org/91/639091/4/check/openstack-tox-py35/a7706d5/ : SUCCESS in 14m 17s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/4/check/openstack-tox-py36/0602c21/ : SUCCESS in 11m 31s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/4/check/openstack-tox-py37/344f05e/ : SUCCESS in 12m 07s\n- openstack-tox-docs http://logs.openstack.org/91/639091/4/check/openstack-tox-docs/cca6140/html/ : SUCCESS in 9m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6f70b8f/ : SUCCESS in 48m 47s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/4/check/devstack-plugin-ceph-tempest/e807628/ : SUCCESS in 1h 36m 09s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/4/check/neutron-grenade-multinode/2e56903/ : SUCCESS in 1h 17m 06s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/4/check/nova-grenade-live-migration/328bfc5/ : SUCCESS in 58m 30s\n- nova-live-migration http://logs.openstack.org/91/639091/4/check/nova-live-migration/64852ba/ : SUCCESS in 51m 09s\n- nova-lvm http://logs.openstack.org/91/639091/4/check/nova-lvm/fb0a253/ : SUCCESS in 57m 06s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/4/check/nova-next/b9c2cb7/ : SUCCESS in 1h 26m 45s\n- nova-tox-functional http://logs.openstack.org/91/639091/4/check/nova-tox-functional/f8e03fd/ : SUCCESS in 17m 34s\n- nova-tox-functional-py35 http://logs.openstack.org/91/639091/4/check/nova-tox-functional-py35/bcf6356/ : SUCCESS in 18m 29s\n- tempest-multinode-full http://logs.openstack.org/91/639091/4/check/tempest-multinode-full/aa3a04b/ : SUCCESS in 2h 13m 31s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/91/639091/4/check/tempest-slow-py3/e0a433c/ : SUCCESS in 2h 18m 51s","accounts_in_message":[],"_revision_number":4},{"id":"c4a97fc2eab3ecee1fcdc9b1c9eeb296e61551d5","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-03-19 15:31:38.000000000","message":"Patch Set 4:\n\nHeh, FWIW I pretty much always prefer the opposite :) I don\u0027t see any penalty to indirection, and in my experience long methods almost always grow longer as new people come along and dump more stuff in them ...  Shorter methods are not just more readable, but lend themselves to easier testing, easier refactoring, easier documentation, better variable scoping, fewer bugs etc. etc.\n\nIf you can tolerate some very simple Ruby, I thought this talk was fantastic so you might enjoy it too:\n\nhttps://www.youtube.com/watch?v\u003d8bZh5LMaSmE\n\nBut anyway, it\u0027s your call here ;)","accounts_in_message":[],"_revision_number":4},{"id":"06e72e41dcb5276efc7b6ba0a6061d014c513ec4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-19 17:19:58.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/748752f : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a9736f42df8480f9870ad085415291a4df55c14e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-19 17:22:46.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-91-639091-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"deebc5c9dccfa0865ac58c2fe9ddc9ffdf24d324","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-25 23:37:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"059cf2cdbc540d1a4734390c78417e41a87a4b49","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-17 14:57:44.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"78d851525518f9e0bc8ef909747f7680a34a2eac","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 11:46:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7aedff6eccb6df31dee3b4498588024d491c7439","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 12:14:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6dddeb93db8dcbddf00486ebcd76585efd2d1f4f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 12:15:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"13dd5b2251fa98b1e1b3678e32d3e1eef5072dc4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 12:17:04.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/5/check/check-dsvm-tempest-vz7-exe-minimal/6f2fc80 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/5/check/check-dsvm-tempest-vz7-exe-multinode/b4fa6b0 : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"3b484d7cca1be18056616f817ab7379a50f99468","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 12:19:13.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ac29766 : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":5},{"id":"0b8562e9dd4312b772d40bdb7a887fb2b287d281","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-29 12:38:25.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/639091/5 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"b12ff998e2de300eaedee3075cd1f49d4fcd15d6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-29 13:08:53.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31229 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":5},{"id":"81d6e05c25641c8af4a3e1713c4f133a40a6475d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 13:17:42.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/5/check-vote/ext-nova-zuul/2af316f : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":5},{"id":"126790a4b1f1c04b36a80af4d44088944f807237","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 13:54:19.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/tempest-dsvm-full-xenial/9ad5aef/ : SUCCESS in 1h 18m 01s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/tempest-dsvm-full-xenial-py3/9488a30/ : SUCCESS in 1h 39m 02s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/grenade-dsvm-xenial/20f4618/ : SUCCESS in 1h 03m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dc944e95187848ed5e7ce8944cc27f9c147b6505","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:05:53.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/5/check/nova-out-of-tree-pvm/7daa0b9 : FAILURE in 1h 49m 02s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/5/check/nova-in-tree-pvm/3d87a03 : FAILURE in 1h 50m 36s","accounts_in_message":[],"_revision_number":5},{"id":"94225de81faa7f6fc306f2d1b2ebf59d1557b962","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:28:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/639091/5/check/grenade-py3/29a5dc8/ : SUCCESS in 54m 52s\n- tempest-full-py3 http://logs.openstack.org/91/639091/5/check/tempest-full-py3/8c6c24e/ : SUCCESS in 1h 56m 34s\n- openstack-tox-cover http://logs.openstack.org/91/639091/5/check/openstack-tox-cover/6c8aeb8/cover/ : SUCCESS in 21m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/5/check/openstack-tox-lower-constraints/a45469f/ : SUCCESS in 29m 05s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/5/check/openstack-tox-pep8/e34c46a/ : SUCCESS in 13m 34s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/5/check/openstack-tox-py27/acf1c8e/ : SUCCESS in 22m 03s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/5/check/openstack-tox-py36/14a5e52/ : SUCCESS in 18m 53s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/5/check/openstack-tox-py37/56eef13/ : SUCCESS in 18m 05s\n- openstack-tox-docs http://logs.openstack.org/91/639091/5/check/openstack-tox-docs/cee7efb/html/ : SUCCESS in 6m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d9c7edc/ : SUCCESS in 56m 27s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/5/check/devstack-plugin-ceph-tempest/9fe0cd2/ : SUCCESS in 1h 22m 34s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/5/check/neutron-grenade-multinode/cedff70/ : SUCCESS in 1h 02m 52s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/5/check/nova-grenade-live-migration/d0fbe18/ : SUCCESS in 1h 03m 48s\n- nova-live-migration http://logs.openstack.org/91/639091/5/check/nova-live-migration/af30383/ : SUCCESS in 47m 18s\n- nova-lvm http://logs.openstack.org/91/639091/5/check/nova-lvm/eafbeab/ : SUCCESS in 1h 04m 32s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/5/check/nova-next/9a4931b/ : FAILURE in 1h 29m 38s\n- nova-tox-functional http://logs.openstack.org/91/639091/5/check/nova-tox-functional/56adba5/ : SUCCESS in 22m 27s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/5/check/nova-tox-functional-py36/dbdfae7/ : SUCCESS in 22m 52s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/5/check/tempest-slow-py3/5863b49/ : SUCCESS in 2h 09m 42s","accounts_in_message":[],"_revision_number":5},{"id":"45f80195093b38ceaaecf2a9db4a46022e351dc9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-29 14:29:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/5 : SUCCESS in 2h 14m 19s","accounts_in_message":[],"_revision_number":5},{"id":"160112ea0d3d49486fb204a83e6208d87f0b7438","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 14:52:55.000000000","message":"Patch Set 5:\n\nSeems like volume deletion failed before the time-out kicked in:\n\ntempest.api.compute.volumes.test_attach_volume_negative.AttachVolumeNegativeTest.test_attach_attached_volume_to_same_server\n   b\u0027Details: (AttachVolumeNegativeTest:tearDownClass) Failed to delete volume be4ccf72-55c9-4fc9-8875-23eb35fded4b within the required time (196 s)., \u003ctraceback object at 0x7f2dc7200f88\u003e))\u0027","accounts_in_message":[],"_revision_number":5},{"id":"4455259e0e76b01c0453948ff8d4c313fadabd5a","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-29 14:53:00.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a3f21aa9bba31f074797cee35fc655030b9f2b89","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:53:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c5b57f26837e53f16f2a3cb2f9cda6c4701941e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:54:47.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/5/check/check-dsvm-tempest-vz7-exe-minimal/e48188a : FAILURE in 27s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/5/check/check-dsvm-tempest-vz7-exe-multinode/2ea1db0 : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"9fa0c90eabdb71576febf4e1a43e79fd61a487ea","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-29 16:16:28.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31236 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":5},{"id":"7b3413a441101d069439fd078e969cb8b7ee8477","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 16:17:06.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/tempest-dsvm-full-xenial/8e25378/ : SUCCESS in 1h 15m 44s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/tempest-dsvm-full-xenial-py3/46aea2c/ : SUCCESS in 1h 23m 31s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/5/check/grenade-dsvm-xenial/3f956fc/ : SUCCESS in 1h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b7b84e5ba7560efde39ee4771236926b4c63fbe7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-29 17:08:43.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/5/check/nova-out-of-tree-pvm/1028694 : FAILURE in 2h 15m 11s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/5/check/nova-in-tree-pvm/56d909b : SUCCESS in 2h 07m 52s","accounts_in_message":[],"_revision_number":5},{"id":"a3b4aa8c87e5c71eb997af11c552629418c0c34b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 17:17:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/639091/5/check/grenade-py3/0204a7b/ : SUCCESS in 1h 01m 40s\n- tempest-full-py3 http://logs.openstack.org/91/639091/5/check/tempest-full-py3/dd2774e/ : SUCCESS in 1h 27m 58s\n- openstack-tox-cover http://logs.openstack.org/91/639091/5/check/openstack-tox-cover/6367cd2/cover/ : SUCCESS in 14m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/5/check/openstack-tox-lower-constraints/af4e405/ : SUCCESS in 13m 08s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/5/check/openstack-tox-pep8/caef355/ : SUCCESS in 9m 54s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/5/check/openstack-tox-py27/b410c19/ : SUCCESS in 12m 15s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/5/check/openstack-tox-py36/1efa0dc/ : SUCCESS in 11m 51s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/5/check/openstack-tox-py37/4e289b8/ : SUCCESS in 12m 25s\n- openstack-tox-docs http://logs.openstack.org/91/639091/5/check/openstack-tox-docs/9c2af42/html/ : SUCCESS in 5m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/38c8d28/ : SUCCESS in 52m 41s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/5/check/devstack-plugin-ceph-tempest/f4eefbe/ : SUCCESS in 1h 52m 06s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/5/check/neutron-grenade-multinode/ce4562f/ : SUCCESS in 1h 11m 06s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/5/check/nova-grenade-live-migration/8863c87/ : SUCCESS in 1h 06m 44s\n- nova-live-migration http://logs.openstack.org/91/639091/5/check/nova-live-migration/8f2ccb2/ : SUCCESS in 49m 23s\n- nova-lvm http://logs.openstack.org/91/639091/5/check/nova-lvm/bca8d3a/ : SUCCESS in 1h 01m 11s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/5/check/nova-next/a7fd048/ : SUCCESS in 1h 36m 03s\n- nova-tox-functional http://logs.openstack.org/91/639091/5/check/nova-tox-functional/b0a224b/ : SUCCESS in 17m 59s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/5/check/nova-tox-functional-py36/1010305/ : SUCCESS in 17m 13s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/5/check/tempest-slow-py3/c07bdf3/ : SUCCESS in 2h 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"8d41d0b899a35eb9da6c48bd232c2d5d1f8834bc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 17:17:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/91/639091/5/check/EMC_VxFlexOS_NOVA/94253f4/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"1474208af5a24ccbdaf82594f338b3692648933b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-29 17:25:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/5 : SUCCESS in 2h 31m 36s","accounts_in_message":[],"_revision_number":5},{"id":"4ba3338ed7565e819c92d060379b60b51ad61dd9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 17:47:43.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/5/check-vote/ext-nova-zuul/f368389 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":5},{"id":"811fdf3d0e50e0e134b581039729f2b44290047e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-29 20:06:04.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-91-639091-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"bd1ce8800c610e697934f5bbdb4195fc796e413c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-04-30 09:49:38.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)\n\nA few nits and supernits in the commit and I wonder if we could move the save_and_reraise_exception call higher. However I\u0027m -1 on this for  now as there are no specific tests for this case where MIN_LIBVIRT_BETTER_SIGKILL_HANDLING is and isn\u0027t met.","accounts_in_message":[],"_revision_number":5},{"id":"c92f9c877e1d06929554e8c0fed4c972aaaeaa60","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-30 11:03:12.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nI\u0027ll work out a test, given the amount of mocking happening here.  And moving the \u0027save_and_reraise_exception\u0027 makes sense.  Thanks for looking!","accounts_in_message":[],"_revision_number":5},{"id":"1be70df78b1c7f2c73357ee8198bc8ff83efb6a5","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-30 11:50:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"91d30db1b8c4827e195c280fb6683ce5c8ee02c3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 11:50:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6dc5678cfe24a6ad6767f25d501df4cee9a2cbcc","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 11:51:41.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/6/check/check-dsvm-tempest-vz7-exe-minimal/9205474 : FAILURE in 29s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/6/check/check-dsvm-tempest-vz7-exe-multinode/77b6e61 : FAILURE in 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"961206360244a233875d3c475f81646550c90de4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 11:55:20.000000000","message":"Patch Set 6:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d68a960 : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":6},{"id":"3260f9031635591c4c0ef97ecc3cde8126f89df8","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-30 12:14:19.000000000","message":"Patch Set 6:\n\n* pci-test http://52.27.155.124/pci/639091/6 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"65578de38485b61d290dff36edd3621ccdd3fc38","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-30 13:02:35.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31296 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9f1771f1a046860bfc6684e8b4f3ac4fe044299f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 13:16:05.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/tempest-dsvm-full-xenial/20e1922/ : SUCCESS in 1h 11m 48s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/tempest-dsvm-full-xenial-py3/6a024d5/ : SUCCESS in 1h 25m 05s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/grenade-dsvm-xenial/dc7c405/ : SUCCESS in 1h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc12af673dec9b5bb1d0d196764e65d07e1ce185","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-30 13:49:26.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/6/check/nova-out-of-tree-pvm/e1d9da4 : FAILURE in 1h 58m 36s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/6/check/nova-in-tree-pvm/f861118 : SUCCESS in 1h 48m 41s","accounts_in_message":[],"_revision_number":6},{"id":"ced4d2db70bb26581fcdcfdf79a113c1129bbdda","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 14:01:22.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/6/check-vote/ext-nova-zuul/45522fa : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":6},{"id":"b4cec11f193a776ca44bf6ed9bf9068d9f7b7cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 14:03:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/639091/6/check/grenade-py3/ba3539b/ : FAILURE in 54m 33s\n- tempest-full-py3 http://logs.openstack.org/91/639091/6/check/tempest-full-py3/58ea8ef/ : SUCCESS in 1h 46m 24s\n- openstack-tox-cover http://logs.openstack.org/91/639091/6/check/openstack-tox-cover/8d1e773/cover/ : SUCCESS in 24m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/6/check/openstack-tox-lower-constraints/3293d9b/ : SUCCESS in 32m 46s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/6/check/openstack-tox-pep8/d10fc89/ : SUCCESS in 10m 54s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/6/check/openstack-tox-py27/803b4e5/ : SUCCESS in 22m 25s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/6/check/openstack-tox-py36/71f9ff1/ : SUCCESS in 27m 49s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/6/check/openstack-tox-py37/e40a277/ : SUCCESS in 29m 12s\n- openstack-tox-docs http://logs.openstack.org/91/639091/6/check/openstack-tox-docs/82c3d94/html/ : SUCCESS in 5m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/78be851/ : SUCCESS in 1h 00m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/6/check/devstack-plugin-ceph-tempest/d2fc0c9/ : SUCCESS in 1h 55m 27s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/6/check/neutron-grenade-multinode/a5174ee/ : SUCCESS in 1h 04m 26s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/6/check/nova-grenade-live-migration/ece3b55/ : SUCCESS in 1h 28m 23s\n- nova-live-migration http://logs.openstack.org/91/639091/6/check/nova-live-migration/8478203/ : SUCCESS in 58m 19s\n- nova-lvm http://logs.openstack.org/91/639091/6/check/nova-lvm/e2db553/ : SUCCESS in 1h 27m 08s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/6/check/nova-next/60b718d/ : SUCCESS in 1h 22m 47s\n- nova-tox-functional http://logs.openstack.org/91/639091/6/check/nova-tox-functional/e9626f5/ : SUCCESS in 37m 43s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/6/check/nova-tox-functional-py36/8d3f885/ : SUCCESS in 34m 58s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/6/check/tempest-slow-py3/3cc5e03/ : SUCCESS in 2h 11m 08s","accounts_in_message":[],"_revision_number":6},{"id":"ce918bc3cd4bda1beb15896052c7899137e944e6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-30 14:05:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/6 : SUCCESS in 2h 14m 37s","accounts_in_message":[],"_revision_number":6},{"id":"b3dcb3e46701934633f7553b66889c4ce46b6672","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 17:56:25.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e54bf61102eaf29e75c31f842978b6355862949e","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-30 21:06:15.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"75b27018407ba5057c9c1a0c8569f2e9769e5a9a","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-04-30 21:06:27.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"0eb9644f2322250e9db2c9d7e87bb43dca590aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 00:18:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/639091/6/check/grenade-py3/140716d/ : SUCCESS in 1h 09m 38s\n- tempest-full-py3 http://logs.openstack.org/91/639091/6/check/tempest-full-py3/a99f71c/ : SUCCESS in 1h 23m 11s\n- openstack-tox-cover http://logs.openstack.org/91/639091/6/check/openstack-tox-cover/411f3be/cover/ : SUCCESS in 17m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/6/check/openstack-tox-lower-constraints/6314f69/ : SUCCESS in 13m 14s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/6/check/openstack-tox-pep8/40c29a9/ : SUCCESS in 10m 35s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/6/check/openstack-tox-py27/e7cd123/ : SUCCESS in 12m 11s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/6/check/openstack-tox-py36/5e188c3/ : SUCCESS in 14m 54s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/6/check/openstack-tox-py37/fee208d/ : SUCCESS in 14m 15s\n- openstack-tox-docs http://logs.openstack.org/91/639091/6/check/openstack-tox-docs/b3217e6/html/ : SUCCESS in 6m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0717ae3/ : SUCCESS in 52m 43s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/6/check/devstack-plugin-ceph-tempest/05a25f4/ : SUCCESS in 1h 39m 34s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/6/check/neutron-grenade-multinode/49f113a/ : SUCCESS in 1h 06m 13s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/6/check/nova-grenade-live-migration/3cae616/ : SUCCESS in 1h 08m 36s\n- nova-live-migration http://logs.openstack.org/91/639091/6/check/nova-live-migration/cdb7b03/ : SUCCESS in 54m 47s\n- nova-lvm http://logs.openstack.org/91/639091/6/check/nova-lvm/dc3250e/ : SUCCESS in 58m 35s (non-voting)\n- nova-multi-cell http://logs.openstack.org/91/639091/6/check/nova-multi-cell/9fa01b7/ : FAILURE in 1h 57m 10s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/6/check/nova-next/5dbb018/ : SUCCESS in 1h 34m 44s\n- nova-tox-functional http://logs.openstack.org/91/639091/6/check/nova-tox-functional/76b251f/ : SUCCESS in 17m 09s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/6/check/nova-tox-functional-py36/82a0712/ : SUCCESS in 16m 46s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/6/check/tempest-slow-py3/1ee7085/ : SUCCESS in 2h 10m 59s","accounts_in_message":[],"_revision_number":6},{"id":"200d1fee63ec3e45ecf8754081a8738e9035d260","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 00:52:34.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/91/639091/6/check/EMC_VxFlexOS_NOVA/00487d8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"299179aeac3ed0f9c214a334dde5a607e724edd1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 00:59:12.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/6/check/check-dsvm-tempest-vz7-exe-minimal/2e4fbc1 : FAILURE in 53s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/6/check/check-dsvm-tempest-vz7-exe-multinode/43f5d3a : FAILURE in 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"fd0c6ed7a8b775c70e8065a3ce3693de212f7cd6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-01 01:50:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e60b1366c88b75137710237014b70524a5723d1a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-01 03:26:49.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/6/check/nova-out-of-tree-pvm/df376f0 : FAILURE in 2h 19m 36s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/6/check/nova-in-tree-pvm/83f1fda : FAILURE in 2h 22m 20s","accounts_in_message":[],"_revision_number":6},{"id":"33aa74ae3e46524deefa9a4712c32757756b9010","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-01 11:38:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/6 : SUCCESS in 2h 20m 14s","accounts_in_message":[],"_revision_number":6},{"id":"5c605033bfff5a55fc3952f920b4a2b30d02d393","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-01 12:40:26.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-91-639091-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"7970e735154361eb4151e0b090bced63884878d3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-05-01 13:41:58.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31367 : FAILURE in 1h 42m 58s","accounts_in_message":[],"_revision_number":6},{"id":"ceda3ad85c4d15c7519d1e399f3c59722cfb30f0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-01 14:08:15.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/91/639091/6/check-vote/ext-nova-zuul/45699f0 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":6},{"id":"05857fea3e2f3bb78f61c394e1fa2e6252ef7d8d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-01 17:20:44.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/tempest-dsvm-full-xenial/a02eacc/ : SUCCESS in 1h 32m 43s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/tempest-dsvm-full-xenial-py3/556a1af/ : SUCCESS in 1h 29m 27s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/6/check/grenade-dsvm-xenial/92a8f3e/ : SUCCESS in 1h 07m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cac767dc4f3fa8e32df47cd70af3d11e53a45f95","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 15:36:30.000000000","message":"Patch Set 6: Code-Review-1\n\nThis looks pretty good but yeah, we need that missing test case before we can go any further with this","accounts_in_message":[],"_revision_number":6},{"id":"fdda800d5c2ca9d20151bd4ce229cb0e8ab6c54d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-05-27 15:42:22.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThe fix LGTM but this still needs tests.","accounts_in_message":[],"_revision_number":6},{"id":"99b025a4102cd4cf71d3067c3bba08743e7ad65d","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-28 15:55:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7384e63cebf147ea3c26d3e2c4bd7c8fdcfaee4","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 09:52:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2279a109f62d04e7f19dd1235f29c28413f2adaa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 09:53:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"67c162524d1567649e36849b917fa9e78e31c57b","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 10:02:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3552624833cbf2714aac2ae7509cbde087a4f055","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 10:13:04.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3f893ebd5d8a56bd7b0ecc2b76e5a8ff7345e45f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-29 10:16:01.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/7/check/nova-out-of-tree-pvm/5351e82 : FAILURE in 22m 52s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/7/check/nova-in-tree-pvm/cb3b9e6 : FAILURE in 20m 15s","accounts_in_message":[],"_revision_number":7},{"id":"2c323bffe9017a281611caae63ba985c36ae7f0d","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 10:43:27.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)\n\n-1 only for the references to _destroy().  Other than that it looks fine to me, but I really would love to see that long method refactored: ideally via one refactoring commit which splits a chunk off into _poweroff() or similar, followed by another refactoring commit which splits off the errno.EBUSY case into another _retry_destroy() method, finally followed by this commit which changes the logic.","accounts_in_message":[],"_revision_number":7},{"id":"23db0a6d227212045e866df04d3da7c435046b69","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-29 10:50:54.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nLGTM. I\u0027m +1 once you delete the #FIXME. You could also address Adam\u0027s nits in the commit message.","accounts_in_message":[],"_revision_number":7},{"id":"85a8a6ac609f822c97af380a6481adb3be1fa9f5","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 10:53:21.000000000","message":"Patch Set 7:\n\nOK, just realised that some of my previous comments are wrong, because I missed that a) there is also _wait_for_destroy(), b) this is defined within _destroy() after all the code you are touching, and c) there are 11 lines of code within _destroy() *after* _wait_for_destroy().  This perfectly illustrates my prior point about how bad long methods are for causing confusion :-(\n\nBut hopefully this is enough food for thought that you can figure out a clean way to refactor this.  IMHO the ideal scenario is that the top-level timeout-retry logic should be expressed in a self-contained method of not more than 10-15 lines, and that it calls out to helper methods to do the donkey work.  In contrast, right now I am staring at this code and feeling pretty helplessly confused about how it\u0027s supposed to work.","accounts_in_message":[],"_revision_number":7},{"id":"37ac5215eccbc65c50f80ad8f932511eed056ab0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 11:07:36.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/7/check/check-dsvm-tempest-vz7-exe-minimal/8ae79da : FAILURE in 16m 27s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/7/check/check-dsvm-tempest-vz7-exe-multinode/1b83b83 : FAILURE in 10m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"63964535c4263d8604b72ab02be5f174841911c3","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-29 11:09:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"385624516e8dd89fc2a62e6d718bc25df68f89ea","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-29 11:19:24.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639091/7 : FAILURE in 19m 07s","accounts_in_message":[],"_revision_number":7},{"id":"3fcd0809b002fcaf9503a0dbc535f80b2827d0b2","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 11:37:12.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nFixing Adam\u0027s points.  And thanks to Matt, too, for the help on #openstack-nova channel for the unit test sleuthing; much appreciated.","accounts_in_message":[],"_revision_number":7},{"id":"2a85d7db71eb7a9cebe12e23726c91bce9792fcb","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 11:46:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"157ea006ddd79712cebcca2ac25daa54a2f79e5c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 11:46:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"60ebec452cccdc3969d6dded8b51f31c6362fcb1","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-29 12:08:42.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/8/check/nova-out-of-tree-pvm/58ccb6e : FAILURE in 21m 46s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/8/check/nova-in-tree-pvm/29b0660 : FAILURE in 20m 28s","accounts_in_message":[],"_revision_number":8},{"id":"b31b1ec790416b93991b06031b852fa9d77b8a30","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 12:24:39.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/8/check/check-dsvm-tempest-vz7-exe-minimal/e56a77b : FAILURE in 16m 38s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/8/check/check-dsvm-tempest-vz7-exe-multinode/58be35e : FAILURE in 10m 58s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"7ef08767f4c228ce23c1e011771866506ca9e884","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-29 12:26:11.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)\n\n2 nits, but LGTM.","accounts_in_message":[],"_revision_number":8},{"id":"24c1e8b2eba1f96d38efc7c911621a6490be3482","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 12:51:09.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d497cce2a1c3722c5fadc00c1986e9c152bd9924","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 12:52:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a84052d3d02ae953b88ee8ddfe000d42f8a7a845","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 12:53:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"07fd2993001cd393cd8ce5fef1ca7411d5f5b3b5","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-29 13:04:50.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)\n\nStill LGTM. Different nit ;)","accounts_in_message":[],"_revision_number":9},{"id":"f16ace3732cb9cf5006d3d2ae3c3f6bf907cf659","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 13:06:59.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/9/check/check-dsvm-tempest-vz7-exe-minimal/4d99bb6 : FAILURE in 11m 22s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/9/check/check-dsvm-tempest-vz7-exe-multinode/34fd7fa : FAILURE in 12m 32s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":9},{"id":"89aa9c9802f1da6beb5befe1330ea6b5202a5156","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-29 13:15:00.000000000","message":"Patch Set 9:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/9/check/nova-out-of-tree-pvm/cb06774 : FAILURE in 21m 42s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/9/check/nova-in-tree-pvm/6e78b6e : FAILURE in 20m 53s","accounts_in_message":[],"_revision_number":9},{"id":"1809f868be75b4ef8e90a29a8f2d7574e0aab50e","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 13:34:18.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)\n\nLGTM; -1 only for the TODO.  Even though the \\d+ tickles my OCD since it\u0027s a string unnecessarily impersonating a regex, it\u0027s not really enough to deserve a -1.","accounts_in_message":[],"_revision_number":9},{"id":"25f218464eacbd13c9e2acfcf98252d26493e3e0","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 13:34:50.000000000","message":"Patch Set 9:\n\nOh, and I\u0027ll gratefully review the discussed refactoring when it arrives ;-)","accounts_in_message":[],"_revision_number":9},{"id":"ffc6942637f4b25e1e696bbcfeb957b0522416ec","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-29 14:32:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"13a0e8564fa97883fa7d33e84424238067aa5fe0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-05-29 14:45:10.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-32288 : SUCCESS in 1h 51m 56s","accounts_in_message":[],"_revision_number":9},{"id":"196a0854cc4507cd43a3560e43de45543e7897eb","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 15:11:15.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c3af7a298ddecd5a64c2c0a5f69635d7640c1e9f","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 15:21:59.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"999636e352b2c29044cf71ab0cc4d7ef66db66a0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 15:22:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8e1c50b530d9c711b5e71d3010779cba4c9c2c5b","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 15:29:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"568c6f86a9a4ab22eaa531a3fea8076b2064eb98","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 15:34:00.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/10/check/check-dsvm-tempest-vz7-exe-minimal/24241f8 : FAILURE in 10m 46s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/10/check/check-dsvm-tempest-vz7-exe-multinode/b0b746b : FAILURE in 10m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":10},{"id":"24dc81351181da99a58a39772fd67be885f1276f","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-05-29 15:37:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"084d1ef834fd84b70f7aa67ec9fcef2493435cbd","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 15:38:09.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fd7d62241f04a4b7485b9006f08b592e34025e8b","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-05-29 15:39:34.000000000","message":"Patch Set 11: Code-Review+1\n\nAll micro-nits addressed, thanks :)","accounts_in_message":[],"_revision_number":11},{"id":"11f8b2b158d8691ce022aa1b05b8b41143f36019","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 15:50:36.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/11/check/check-dsvm-tempest-vz7-exe-minimal/2327a16 : FAILURE in 2m 02s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/91/639091/11/check/check-dsvm-tempest-vz7-exe-multinode/e6c88d0 : FAILURE in 11m 43s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"84aff1f7d52cfb381fb34335d5725b6affc902ad","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-29 16:00:41.000000000","message":"Patch Set 11:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/91/639091/11/check/nova-out-of-tree-pvm/0de340a : FAILURE in 22m 18s\n- nova-in-tree-pvm http://184.172.12.213/91/639091/11/check/nova-in-tree-pvm/60827fd : FAILURE in 21m 06s","accounts_in_message":[],"_revision_number":11},{"id":"82e59e96b0d572228a7581d1a709c759079a96b4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-05-29 17:09:36.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-32291 : SUCCESS in 1h 31m 31s","accounts_in_message":[],"_revision_number":11},{"id":"f8af3aa8fcb72b82e14d16424f20fec02ab7ee34","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-29 18:03:50.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639091/11 : SUCCESS in 2h 25m 16s","accounts_in_message":[],"_revision_number":11},{"id":"7540f51e5d48342fc58b59b208c43b39e8b6f978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 20:44:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/639091/11/check/grenade-py3/1554c45/ : SUCCESS in 48m 55s\n- tempest-full-py3 http://logs.openstack.org/91/639091/11/check/tempest-full-py3/f885d25/ : SUCCESS in 1h 36m 55s\n- openstack-tox-cover http://logs.openstack.org/91/639091/11/check/openstack-tox-cover/ee42497/cover/ : SUCCESS in 16m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/11/check/openstack-tox-lower-constraints/8afd2e2/ : SUCCESS in 14m 26s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/11/check/openstack-tox-pep8/e4d13bd/ : SUCCESS in 11m 07s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/11/check/openstack-tox-py27/34e63d6/ : SUCCESS in 11m 06s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/11/check/openstack-tox-py36/d7b8668/ : SUCCESS in 11m 39s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/11/check/openstack-tox-py37/4fcf442/ : SUCCESS in 12m 07s\n- openstack-tox-docs http://logs.openstack.org/91/639091/11/check/openstack-tox-docs/d22fc9a/html/ : SUCCESS in 5m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/91/639091/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d585c2d/ : SUCCESS in 49m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/91/639091/11/check/devstack-plugin-ceph-tempest/a21fb29/ : FAILURE in 1h 34m 55s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/91/639091/11/check/neutron-grenade-multinode/346faed/ : SUCCESS in 1h 06m 44s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/11/check/nova-grenade-live-migration/9feba0a/ : SUCCESS in 1h 13m 45s\n- nova-live-migration http://logs.openstack.org/91/639091/11/check/nova-live-migration/cdb451e/ : SUCCESS in 47m 40s\n- nova-lvm http://logs.openstack.org/91/639091/11/check/nova-lvm/45687c3/ : SUCCESS in 1h 00m 03s (non-voting)\n- nova-multi-cell http://logs.openstack.org/91/639091/11/check/nova-multi-cell/bc9c0a8/ : SUCCESS in 1h 42m 06s (non-voting)\n- nova-next http://logs.openstack.org/91/639091/11/check/nova-next/15bf68c/ : SUCCESS in 1h 38m 39s\n- nova-tox-functional http://logs.openstack.org/91/639091/11/check/nova-tox-functional/1a9741d/ : SUCCESS in 20m 48s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/11/check/nova-tox-functional-py36/acaf315/ : SUCCESS in 17m 59s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/11/check/tempest-slow-py3/3d85870/ : SUCCESS in 2h 31m 21s","accounts_in_message":[],"_revision_number":11},{"id":"fd757df8e56871a31f8f24cb8fc3b50742c261d3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-29 20:44:37.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/91/639091/11/check/EMC_VxFlexOS_NOVA/d4a4cda/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"360274e8c9b475d528be8a3d77a389cec20f6d1e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-29 21:47:54.000000000","message":"Patch Set 11: Code-Review+2\n\nThanks, this lgtm now.","accounts_in_message":[],"_revision_number":11},{"id":"578418e7940e138b16033b00dc6429282dca0c04","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-29 21:52:22.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-91-639091-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"dc163d9ddce8bc0d57c7d147e3b6d78eaba1cde4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-30 01:38:55.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/11/check/tempest-dsvm-full-xenial/cabd087/ : SUCCESS in 1h 28m 53s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/11/check/tempest-dsvm-full-xenial-py3/36a8d87/ : SUCCESS in 1h 26m 51s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/91/639091/11/check/grenade-dsvm-xenial/5702284/ : SUCCESS in 1h 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0cb0ea70e67a68eafdb2c9ffa1e3103da8dee1c6","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-05-30 03:27:24.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639091/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5182703 : SUCCESS in 1h 51m 12s","accounts_in_message":[],"_revision_number":11},{"id":"8523dfdd56eceaaba974c24333ad0691585be810","author":{"_account_id":25314,"name":"黄海滔","email":"huanght@awcloud.com","username":"huanght"},"date":"2019-05-30 06:14:45.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"4bc5311e55f6bcdacc06a8d8424d61f71076bc24","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-05-30 08:35:13.000000000","message":"Patch Set 11: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":11},{"id":"012beaec0d094cf1a4ae3c620983539c49d74fb5","author":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"date":"2019-05-31 12:35:29.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"8276a47462689666b8ea4b40720f75bfeab0ae44","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-31 12:54:35.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nYup, fine by me","accounts_in_message":[],"_revision_number":11},{"id":"5d3d7d579602ea3e474c92f2d9e0fb9a2105de7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:55:11.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"18e1b2c6dd9d7ce10950dbb98a859186fb6c58fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 15:41:23.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/639091/11/gate/grenade-py3/4e7c643/ : SUCCESS in 1h 02m 40s\n- tempest-full-py3 http://logs.openstack.org/91/639091/11/gate/tempest-full-py3/3216af5/ : SUCCESS in 1h 38m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/639091/11/gate/openstack-tox-lower-constraints/592f77b/ : SUCCESS in 13m 19s\n- openstack-tox-pep8 http://logs.openstack.org/91/639091/11/gate/openstack-tox-pep8/15c4ea0/ : SUCCESS in 10m 45s\n- openstack-tox-py27 http://logs.openstack.org/91/639091/11/gate/openstack-tox-py27/a37fc36/ : SUCCESS in 16m 43s\n- openstack-tox-py36 http://logs.openstack.org/91/639091/11/gate/openstack-tox-py36/54cd1d3/ : SUCCESS in 11m 12s\n- openstack-tox-py37 http://logs.openstack.org/91/639091/11/gate/openstack-tox-py37/2a21051/ : SUCCESS in 20m 48s\n- openstack-tox-docs http://logs.openstack.org/91/639091/11/gate/openstack-tox-docs/2e887e9/html/ : SUCCESS in 9m 21s\n- nova-grenade-live-migration http://logs.openstack.org/91/639091/11/gate/nova-grenade-live-migration/a23b197/ : SUCCESS in 1h 05m 16s\n- nova-live-migration http://logs.openstack.org/91/639091/11/gate/nova-live-migration/b143838/ : SUCCESS in 1h 15m 17s\n- nova-tox-functional http://logs.openstack.org/91/639091/11/gate/nova-tox-functional/85cc9cc/ : SUCCESS in 24m 04s\n- nova-tox-functional-py36 http://logs.openstack.org/91/639091/11/gate/nova-tox-functional-py36/16080d3/ : SUCCESS in 20m 59s\n- nova-next http://logs.openstack.org/91/639091/11/gate/nova-next/c987ec1/ : SUCCESS in 1h 22m 12s\n- tempest-slow-py3 http://logs.openstack.org/91/639091/11/gate/tempest-slow-py3/a0bd3c1/ : SUCCESS in 2h 14m 42s","accounts_in_message":[],"_revision_number":11},{"id":"100120113fa48132a0448df8a5c417fad806c7ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 15:41:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"e65990f6bcc82c3648013a2c13fdd5978e2cd86d","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-06-25 15:51:09.000000000","message":"Patch Set 11: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 118cf0c5920b5ff41333fd304a934dd7e5f4e1a8","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"10d50ca4e210039aeae84cb9bd5d18895948af54","revisions":{"cb4bbe582760479ec90aef8abed13b4a37130791":{"kind":"REWORK","_number":1,"created":"2019-02-25 13:39:08.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/1"}}},"commit":{"parents":[{"commit":"a90c8e1a359a236e06f3a78df74f55808bbef31b","subject":"Merge \"Remove misleading code from _move_operation_alloc_request()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a90c8e1a359a236e06f3a78df74f55808bbef31b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 13:36:36.000000000","tz":60},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nNova commit 3907867 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroY()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large Compute nodes (3TB of RAM) with high instance\n   add or delete churn, where the current timout may not be sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb4bbe582760479ec90aef8abed13b4a37130791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb4bbe582760479ec90aef8abed13b4a37130791"}]},"branch":"refs/heads/master"},"a8e625e412438a06b4370032a8cbc7ef0880b863":{"kind":"REWORK","_number":2,"created":"2019-02-25 15:10:10.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/2"}}},"commit":{"parents":[{"commit":"a90c8e1a359a236e06f3a78df74f55808bbef31b","subject":"Merge \"Remove misleading code from _move_operation_alloc_request()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a90c8e1a359a236e06f3a78df74f55808bbef31b"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 15:08:17.000000000","tz":60},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nNova commit 3907867 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroY()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large Compute nodes (3TB of RAM) with high instance\n   add or delete churn, where the current timout may not be sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8e625e412438a06b4370032a8cbc7ef0880b863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8e625e412438a06b4370032a8cbc7ef0880b863"}]},"branch":"refs/heads/master"},"a8f97b78b789f747a919c230e83d2c3eb9ce7f6d":{"kind":"REWORK","_number":3,"created":"2019-03-08 13:30:55.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/3"}}},"commit":{"parents":[{"commit":"dc1ef24fac893c53dd8a38262050cec658b5a8ce","subject":"Merge \"Move additional IP address management to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc1ef24fac893c53dd8a38262050cec658b5a8ce"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-08 13:27:18.000000000","tz":60},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nNova commit 3907867 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroY()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large Compute nodes (3TB of RAM) with high instance\n   creation or delete churn, where the current timout may not be\n   sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8f97b78b789f747a919c230e83d2c3eb9ce7f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8f97b78b789f747a919c230e83d2c3eb9ce7f6d"}]},"branch":"refs/heads/master"},"e3c3e56a36f410870598ac8f7f9c976621e6f5df":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-19 12:49:41.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/4"}}},"commit":{"parents":[{"commit":"f58f73978e3c533304b06a4d1eb6abe3f229bc10","subject":"Merge \"Remove stale aggregates notes from scheduler evolution doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f58f73978e3c533304b06a4d1eb6abe3f229bc10"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-03-19 12:43:26.000000000","tz":60},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nNova commit 3907867 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroY()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large Compute nodes (3TB of RAM) with high instance\n   creation or delete churn, where the current timout may not be\n   sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3c3e56a36f410870598ac8f7f9c976621e6f5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3c3e56a36f410870598ac8f7f9c976621e6f5df"}]},"branch":"refs/heads/master"},"362045bf2bd9d9a34d7307ce0c07aec0c325cfbb":{"kind":"REWORK","_number":5,"created":"2019-04-29 12:14:31.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/5"}}},"commit":{"parents":[{"commit":"e25d59078e61fe9f925dbef53dfe88e575d34dab","subject":"Merge \"Add minimum value in max_concurrent_live_migrations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e25d59078e61fe9f925dbef53dfe88e575d34dab"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-29 11:49:24.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nNova commit 3907867 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroY()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large Compute nodes (3TB of RAM) with high instance\n   creation or delete churn, where the current timout may not be\n   sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/362045bf2bd9d9a34d7307ce0c07aec0c325cfbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/362045bf2bd9d9a34d7307ce0c07aec0c325cfbb"}]},"branch":"refs/heads/master"},"dc6ca1110fceec036e2b2056e45a19c367005f70":{"kind":"REWORK","_number":6,"created":"2019-04-30 11:50:04.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/6"}}},"commit":{"parents":[{"commit":"50c5fa3f4d977950fb478fdff78956898aeec60e","subject":"Merge \"Delete require_instance_exists_using_uuid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/50c5fa3f4d977950fb478fdff78956898aeec60e"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-04-30 11:41:56.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIGKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large compute nodes with high instance creation or\n   delete churn, where the current timout may not be sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc6ca1110fceec036e2b2056e45a19c367005f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc6ca1110fceec036e2b2056e45a19c367005f70"}]},"branch":"refs/heads/master"},"dd5b19b3f90ed91154d4ef5dacec06f8f8b930be":{"kind":"REWORK","_number":7,"created":"2019-05-29 09:52:38.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/7"}}},"commit":{"parents":[{"commit":"d4f58f5eb6e68d0348868efc45212b70feb0bde1","subject":"Merge \"Fix live-migration when glance image deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4f58f5eb6e68d0348868efc45212b70feb0bde1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-29 09:52:11.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\ndestroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIGKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large compute nodes with high instance creation or\n   delete churn, where the current timout may not be sufficient.\n\n - Retry the destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd5b19b3f90ed91154d4ef5dacec06f8f8b930be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd5b19b3f90ed91154d4ef5dacec06f8f8b930be"}]},"branch":"refs/heads/master"},"9903a0b6f615cbae4c23264a7cd4ef0a7303ff39":{"kind":"REWORK","_number":8,"created":"2019-05-29 11:46:24.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/8"}}},"commit":{"parents":[{"commit":"d4f58f5eb6e68d0348868efc45212b70feb0bde1","subject":"Merge \"Fix live-migration when glance image deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4f58f5eb6e68d0348868efc45212b70feb0bde1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-29 11:45:47.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\n_destroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n - Increment the counter to retry the _destroy() call from 3 to 6, thus\n   increasing the total time from 15 to 30 seconds, before SIGKILL takes\n   effect.  And it matches the (more graceful) behaviour of libvirt\n   v4.7.0.  This also gives breathing room for Nova instances running in\n   environments with large compute nodes with high instance creation or\n   delete churn, where the current timout may not be sufficient.\n\n - Retry the _destroy() API call _only_ if MIN_LIBVIRT_VERSION is\n   lower than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9903a0b6f615cbae4c23264a7cd4ef0a7303ff39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9903a0b6f615cbae4c23264a7cd4ef0a7303ff39"}]},"branch":"refs/heads/master"},"41c7881a635d4b21bb5f0eb8e8a7e0d66c0a4998":{"kind":"REWORK","_number":9,"created":"2019-05-29 12:52:47.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/9"}}},"commit":{"parents":[{"commit":"d4f58f5eb6e68d0348868efc45212b70feb0bde1","subject":"Merge \"Fix live-migration when glance image deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4f58f5eb6e68d0348868efc45212b70feb0bde1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-29 12:48:07.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\n_destroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n  - Increment the counter to retry the _destroy() call from 3 to 6, thus\n    increasing the total time from 15 to 30 seconds, before SIGKILL\n    takes effect.  And it matches the (more graceful) behaviour of\n    libvirt v4.7.0.  This also gives breathing room for Nova instances\n    running in environments with large compute nodes with high instance\n    creation or delete churn, where the current timout may not be\n    sufficient.\n\n  - Retry the _destroy() API call _only_ if MIN_LIBVIRT_VERSION is lower\n    than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/41c7881a635d4b21bb5f0eb8e8a7e0d66c0a4998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/41c7881a635d4b21bb5f0eb8e8a7e0d66c0a4998"}]},"branch":"refs/heads/master"},"d2827e79c50723c044742af66dff8883da4c6a31":{"kind":"REWORK","_number":10,"created":"2019-05-29 15:21:59.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/10"}}},"commit":{"parents":[{"commit":"d4f58f5eb6e68d0348868efc45212b70feb0bde1","subject":"Merge \"Fix live-migration when glance image deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4f58f5eb6e68d0348868efc45212b70feb0bde1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-29 15:18:56.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\n_destroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n  - Increment the counter to retry the _destroy() call from 3 to 6, thus\n    increasing the total time from 15 to 30 seconds, before SIGKILL\n    takes effect.  And it matches the (more graceful) behaviour of\n    libvirt v4.7.0.  This also gives breathing room for Nova instances\n    running in environments with large compute nodes with high instance\n    creation or delete churn, where the current timout may not be\n    sufficient.\n\n  - Retry the _destroy() API call _only_ if MIN_LIBVIRT_VERSION is lower\n    than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2827e79c50723c044742af66dff8883da4c6a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2827e79c50723c044742af66dff8883da4c6a31"}]},"branch":"refs/heads/master"},"10d50ca4e210039aeae84cb9bd5d18895948af54":{"kind":"REWORK","_number":11,"created":"2019-05-29 15:37:38.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/91/639091/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/91/639091/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/91/639091/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/91/639091/11"}}},"commit":{"parents":[{"commit":"d4f58f5eb6e68d0348868efc45212b70feb0bde1","subject":"Merge \"Fix live-migration when glance image deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4f58f5eb6e68d0348868efc45212b70feb0bde1"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-02-25 12:26:24.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2019-05-29 15:35:17.000000000","tz":120},"subject":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path","message":"libvirt: Rework \u0027EBUSY\u0027 (SIGKILL) error handling code path\n\nChange ID I128bf6b939 (libvirt: handle code\u003d38 + sigkill (ebusy) in\n_destroy()) handled the case where a QEMU process \"refuses to die\" within\na given timeout period set by libvirt.\n\nOriginally, libvirt sent SIGTERM (allowing the process to clean-up\nresources), then waited 10 seconds, if the guest didn\u0027t go away.  Then\nit sent, the more lethal, SIGKILL and waited another 5 seconds for it to\ntake effect.\n\nFrom libvirt v4.7.0 onwards, libvirt increased[1][2] the time it waits\nfor a guest hard shutdown to complete.  It now waits for 30 seconds for\nSIGKILL to work (instead of 5).  Also, additional wait time is added if\nthere are assigned PCI devices, as some of those tend to slow things\ndown.\n\nIn this change:\n\n  - Increment the counter to retry the _destroy() call from 3 to 6, thus\n    increasing the total time from 15 to 30 seconds, before SIGKILL\n    takes effect.  And it matches the (more graceful) behaviour of\n    libvirt v4.7.0.  This also gives breathing room for Nova instances\n    running in environments with large compute nodes with high instance\n    creation or delete churn, where the current timout may not be\n    sufficient.\n\n  - Retry the _destroy() API call _only_ if MIN_LIBVIRT_VERSION is lower\n    than 4.7.0.\n\n[1] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommitdiff;h\u003d9a4e4b9\n    (process: wait longer 5-\u003e30s on hard shutdown)\n[2] https://libvirt.org/git/?p\u003dlibvirt.git;a\u003dcommit;h\u003dbe2ca04 (\"process:\n    wait longer on kill per assigned Hostdev\")\n\nRelated-bug: #1353939\n\nChange-Id: If2035cac931c42c440d61ba97ebc7e9e92141a28\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10d50ca4e210039aeae84cb9bd5d18895948af54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10d50ca4e210039aeae84cb9bd5d18895948af54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
