)]}'
{"id":"openstack%2Fnova~639281","triplet_id":"openstack%2Fnova~master~I875483c869a253648fd90a9d3510f79cee509ce5","project":"openstack/nova","branch":"master","topic":"my-own-personal-alternative-universe","hashtags":[],"change_id":"I875483c869a253648fd90a9d3510f79cee509ce5","subject":"Move router advertisement daemon restarts to privsep.","status":"MERGED","created":"2019-02-26 09:43:48.000000000","updated":"2019-08-20 04:56:59.000000000","submitted":"2019-08-20 04:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639281-1566277019419-ac68746e","meta_rev_id":"b487e22fff76d59d43279ae5f301094f745fd1f3","_number":639281,"virtual_id_number":639281,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 21:58:11.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-16 10:30:32.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-16 12:57:22.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 22:57:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-08-16 11:21:05.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-08-19 20:20:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-08-20 04:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 00:00:27.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-08-19 22:32:01.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-08-19 18:18:07.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"date":"2019-08-19 18:19:48.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-16 10:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-16 10:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-16 10:27:37.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2019-08-16 10:27:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-16 10:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-06 07:40:18.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-03-06 19:09:40.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-03-21 17:46:46.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-03-27 09:47:00.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2019-04-01 06:44:59.000000000","updated_by":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"reviewer":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"state":"REVIEWER"},{"updated":"2019-06-26 19:59:44.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 01:27:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 06:28:23.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 10:30:32.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 10:39:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-08-16 11:21:05.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 12:57:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 18:18:07.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-19 18:19:48.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 20:20:32.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-08-19 21:58:11.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 22:32:01.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-19 22:57:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 00:00:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 04:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b5e3dde4b8732a0a007fd7313278234a99eab3e","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-02-26 09:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b393f83d0a7a232f03055c3223425c92850b28a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 10:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f813985965f2778bfcbad93284ec59e30bde469c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-26 12:32:22.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28269 : FAILURE in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9ba2b9df107d17c6e4c372c01172a458654cf89a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 12:45:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/1/check/grenade-py3/1441ff0/ : SUCCESS in 58m 09s\n- tempest-full-py3 http://logs.openstack.org/81/639281/1/check/tempest-full-py3/acee381/ : SUCCESS in 1h 31m 00s\n- openstack-tox-cover http://logs.openstack.org/81/639281/1/check/openstack-tox-cover/a4d76fe/cover/ : SUCCESS in 19m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/1/check/openstack-tox-lower-constraints/2319218/ : SUCCESS in 19m 05s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/1/check/openstack-tox-pep8/e7e2229/ : SUCCESS in 10m 48s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/1/check/openstack-tox-py27/ec1ec67/ : SUCCESS in 13m 49s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/1/check/openstack-tox-py35/ea8a12d/ : SUCCESS in 15m 10s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/1/check/openstack-tox-py36/71de42b/ : SUCCESS in 15m 39s\n- openstack-tox-docs http://logs.openstack.org/81/639281/1/check/openstack-tox-docs/42b3855/html/ : SUCCESS in 6m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/10af4b8/ : SUCCESS in 53m 59s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/1/check/devstack-plugin-ceph-tempest/80f560d/ : SUCCESS in 1h 49m 28s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639281/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/3d6b880/ : SUCCESS in 1h 05m 36s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/1/check/neutron-grenade-multinode/f45d6a6/ : SUCCESS in 1h 24m 01s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/1/check/neutron-tempest-linuxbridge/f66420e/ : SUCCESS in 1h 31m 33s\n- nova-live-migration http://logs.openstack.org/81/639281/1/check/nova-live-migration/11d00aa/ : SUCCESS in 49m 35s\n- nova-lvm http://logs.openstack.org/81/639281/1/check/nova-lvm/f63796f/ : SUCCESS in 58m 36s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/1/check/nova-next/27e8ed1/ : SUCCESS in 1h 39m 41s\n- nova-tox-functional http://logs.openstack.org/81/639281/1/check/nova-tox-functional/9fa7aa4/ : SUCCESS in 20m 10s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/1/check/nova-tox-functional-py35/23e5253/ : SUCCESS in 18m 40s\n- tempest-multinode-full http://logs.openstack.org/81/639281/1/check/tempest-multinode-full/b8af8d2/ : SUCCESS in 1h 43m 53s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/1/check/tempest-slow-py3/d029685/ : SUCCESS in 2h 09m 54s","accounts_in_message":[],"_revision_number":1},{"id":"991b7bc01cb55f1d5b311e24bde4fe9b02f587de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 12:48:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/1/check/EMC_VxFlexOS_NOVA/401afde/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"bfcbb42e5cdd2b6abfe3c40c7635b0d43b23a74a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-26 13:02:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/1 : FAILURE in 37m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b28fe1c78b671a50f343bd220ceea56d00aa789b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-26 14:50:47.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"10aed64846297bb66ffbb74f6b47baa9afce304f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-26 17:33:50.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/1/check-vote/ext-nova-zuul/8e72b52 : FAILURE in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7d82510b7c4cae7d6ec024350e9a716ad8afc504","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 10:02:10.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d9c3d80 : FAILURE in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fac3f7303e0cd9534c7349dca6eb5f4eab663c87","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-27 15:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/1/check/tempest-dsvm-full-xenial/7e77408/ : SUCCESS in 1h 22m 26s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/1/check/tempest-dsvm-full-xenial-py3/55ddfdd/ : FAILURE in 17m 29s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/1/check/grenade-dsvm-xenial/2972eb2/ : FAILURE in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a822e1aad1b25dae2c95409701dc610048afc7c6","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-05 22:35:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ce9782de5d7000cc40a2b2f4b2b6b7f643319376","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-05 23:05:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8e264b97a48b6fa57da9c6daecf68dc05290c75","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-06 00:06:16.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/2/check/nova-out-of-tree-pvm/afcbc9f : FAILURE in 43m 49s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/2/check/nova-in-tree-pvm/323db48 : FAILURE in 43m 29s","accounts_in_message":[],"_revision_number":2},{"id":"48e3f102d6b17431a8a6757aa5b9e28082b38156","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 03:51:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/2/check/grenade-py3/3d3f6cb/ : SUCCESS in 1h 04m 08s\n- tempest-full-py3 http://logs.openstack.org/81/639281/2/check/tempest-full-py3/994e0f5/ : SUCCESS in 1h 30m 31s\n- openstack-tox-cover http://logs.openstack.org/81/639281/2/check/openstack-tox-cover/a7bedfa/cover/ : SUCCESS in 16m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/2/check/openstack-tox-lower-constraints/976cd94/ : SUCCESS in 13m 21s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/2/check/openstack-tox-pep8/f861259/ : SUCCESS in 10m 13s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/2/check/openstack-tox-py27/cc0af19/ : SUCCESS in 13m 57s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/2/check/openstack-tox-py35/62535cf/ : SUCCESS in 12m 12s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/2/check/openstack-tox-py36/3efa8ca/ : SUCCESS in 12m 42s\n- openstack-tox-docs http://logs.openstack.org/81/639281/2/check/openstack-tox-docs/fde6278/html/ : SUCCESS in 6m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/15bc7eb/ : SUCCESS in 51m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/2/check/devstack-plugin-ceph-tempest/fa2c4bc/ : SUCCESS in 1h 28m 32s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639281/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/dc06c77/ : SUCCESS in 1h 11m 30s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/2/check/neutron-grenade-multinode/21eb844/ : SUCCESS in 1h 17m 33s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/2/check/neutron-tempest-linuxbridge/2ed382c/ : SUCCESS in 1h 34m 51s\n- nova-live-migration http://logs.openstack.org/81/639281/2/check/nova-live-migration/bdff2cd/ : SUCCESS in 50m 32s\n- nova-lvm http://logs.openstack.org/81/639281/2/check/nova-lvm/e29bc1e/ : SUCCESS in 1h 09m 43s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/2/check/nova-next/409c5db/ : SUCCESS in 2h 01m 33s\n- nova-tox-functional http://logs.openstack.org/81/639281/2/check/nova-tox-functional/d4fe4d0/ : SUCCESS in 17m 45s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/2/check/nova-tox-functional-py35/0702000/ : SUCCESS in 17m 51s\n- tempest-multinode-full http://logs.openstack.org/81/639281/2/check/tempest-multinode-full/ce97215/ : SUCCESS in 1h 35m 15s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/2/check/tempest-slow-py3/00c3869/ : SUCCESS in 2h 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"26ad88fb7f66df1b440d79104e5c6a068b25f7e9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-06 03:54:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/2/check/EMC_VxFlexOS_NOVA/9c6521f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"20ebc8f35ae01d93d57fa231a641bc2b5c93006f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 04:49:47.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28738 : SUCCESS in 2h 14m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f0568b649af06f86156e3bd29a9fa9cffe6e80e0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-06 06:04:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/2 : SUCCESS in 2h 21m 17s","accounts_in_message":[],"_revision_number":2},{"id":"abcf2214321710b36e7e4c8fcfb9484577ea1cce","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-06 07:40:18.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"325cf9d19fdcbbffac570f76bbf3d22acddf20ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-06 10:18:59.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"4d54869e3e825afdaea298b7f5175b5f7667bc6b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 13:37:38.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/2/check-vote/ext-nova-zuul/c6dcc39 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":2},{"id":"01169b57566292ba61ae97efee458b6859e8f3de","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-03-06 19:09:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e4706ccc18ac4bef2083da2de231ed9445c70392","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 23:37:31.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28825 : SUCCESS in 2h 30m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ec483215637ac304e6548b4423bd7422fb7d8d73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 00:01:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/3/check/grenade-py3/3c29a0e/ : SUCCESS in 1h 01m 42s\n- tempest-full-py3 http://logs.openstack.org/81/639281/3/check/tempest-full-py3/e17915d/ : SUCCESS in 1h 29m 00s\n- openstack-tox-cover http://logs.openstack.org/81/639281/3/check/openstack-tox-cover/32928aa/cover/ : SUCCESS in 16m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/3/check/openstack-tox-lower-constraints/b3827cd/ : SUCCESS in 14m 50s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/3/check/openstack-tox-pep8/cafe99c/ : SUCCESS in 9m 44s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/3/check/openstack-tox-py27/d332e85/ : SUCCESS in 13m 01s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/3/check/openstack-tox-py35/f4bb9de/ : SUCCESS in 14m 36s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/3/check/openstack-tox-py36/3f32d78/ : SUCCESS in 12m 59s\n- openstack-tox-docs http://logs.openstack.org/81/639281/3/check/openstack-tox-docs/2e09620/html/ : SUCCESS in 7m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/86f238d/ : SUCCESS in 45m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/3/check/devstack-plugin-ceph-tempest/8c26dad/ : FAILURE in 1h 30m 51s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639281/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/f4b3dec/ : SUCCESS in 1h 13m 11s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/3/check/neutron-grenade-multinode/22b44fe/ : SUCCESS in 1h 05m 16s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/3/check/neutron-tempest-linuxbridge/5c07d6b/ : SUCCESS in 1h 33m 34s\n- nova-live-migration http://logs.openstack.org/81/639281/3/check/nova-live-migration/c792237/ : SUCCESS in 46m 26s\n- nova-lvm http://logs.openstack.org/81/639281/3/check/nova-lvm/253c4ca/ : SUCCESS in 1h 07m 18s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/3/check/nova-next/ce15b26/ : SUCCESS in 1h 52m 00s\n- nova-tox-functional http://logs.openstack.org/81/639281/3/check/nova-tox-functional/c737157/ : SUCCESS in 19m 57s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/3/check/nova-tox-functional-py35/54f8214/ : SUCCESS in 21m 15s\n- tempest-multinode-full http://logs.openstack.org/81/639281/3/check/tempest-multinode-full/c7092f1/ : SUCCESS in 1h 47m 21s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/3/check/tempest-slow-py3/bfe4437/ : SUCCESS in 2h 11m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a34064c957044009a9ea505f8915fb05f02ec739","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-07 00:04:40.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/3/check/EMC_VxFlexOS_NOVA/b86b161/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"3687553afaac640323ae1ca8b38ef20ec1bb1d43","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 01:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/3 : SUCCESS in 2h 17m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f5efab018e9b4319b7d5ca5e047a324d2e666f38","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-07 12:06:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"37907631bddbd525ae108d78bec31ab4625cb2f4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-07 14:35:12.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"3841ac39388d43011d9e111eaae502ac6bc5c41a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 14:42:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/4/check/grenade-py3/c4824ef/ : SUCCESS in 58m 21s\n- tempest-full-py3 http://logs.openstack.org/81/639281/4/check/tempest-full-py3/2d7b2b5/ : SUCCESS in 1h 24m 55s\n- openstack-tox-cover http://logs.openstack.org/81/639281/4/check/openstack-tox-cover/24f31ff/ : FAILURE in 14m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/4/check/openstack-tox-lower-constraints/9733590/ : FAILURE in 13m 28s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/4/check/openstack-tox-pep8/110163f/ : FAILURE in 10m 06s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/4/check/openstack-tox-py27/2e8a973/ : FAILURE in 12m 33s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/4/check/openstack-tox-py35/f19993b/ : FAILURE in 12m 43s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/4/check/openstack-tox-py36/a99aed7/ : FAILURE in 10m 51s\n- openstack-tox-docs http://logs.openstack.org/81/639281/4/check/openstack-tox-docs/5790f41/html/ : SUCCESS in 6m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/811cc11/ : SUCCESS in 45m 24s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/4/check/devstack-plugin-ceph-tempest/e3c0103/ : SUCCESS in 1h 32m 55s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639281/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/b437294/ : SUCCESS in 1h 10m 50s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/4/check/neutron-grenade-multinode/875720f/ : FAILURE in 47m 06s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/4/check/neutron-tempest-linuxbridge/e665ea2/ : FAILURE in 1h 10m 01s\n- nova-live-migration http://logs.openstack.org/81/639281/4/check/nova-live-migration/d52351a/ : SUCCESS in 44m 39s\n- nova-lvm http://logs.openstack.org/81/639281/4/check/nova-lvm/8eb455f/ : SUCCESS in 51m 33s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/4/check/nova-next/a7f6171/ : SUCCESS in 1h 53m 02s\n- nova-tox-functional http://logs.openstack.org/81/639281/4/check/nova-tox-functional/aaad866/ : SUCCESS in 19m 53s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/4/check/nova-tox-functional-py35/84359f7/ : SUCCESS in 17m 37s\n- tempest-multinode-full http://logs.openstack.org/81/639281/4/check/tempest-multinode-full/d67fc78/ : SUCCESS in 1h 56m 10s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/4/check/tempest-slow-py3/0512a20/ : SUCCESS in 2h 15m 20s","accounts_in_message":[],"_revision_number":4},{"id":"fde84325e09521c81f8fcf2764f514381771a176","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 14:45:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/4 : SUCCESS in 2h 29m 49s","accounts_in_message":[],"_revision_number":4},{"id":"6094cea96df53a8c2dfaa93e79a9226d6cb4ce56","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-07 15:56:38.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28879 : SUCCESS in 3h 09m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ab925ba861287b43dc084137ed5c3478e65f1114","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-07 22:26:56.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/4/check-vote/ext-nova-zuul/082d87d : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":4},{"id":"20699c78c166e2cc734a0833ffc1f83992575553","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 03:00:48.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/535ef10 : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":4},{"id":"d5e1371d70d799d812f1fb34f5cd35f12b0633e1","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-08 05:17:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a4c6557ac92242f9c71abe17f5a63aadcc5a383c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 07:06:22.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28939 : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":5},{"id":"6db1a3f8642f1627905631ba454b0872b01a3c02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 09:29:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/5/check/grenade-py3/73e23ec/ : SUCCESS in 1h 06m 03s\n- tempest-full-py3 http://logs.openstack.org/81/639281/5/check/tempest-full-py3/28f1a9d/ : SUCCESS in 1h 36m 16s\n- openstack-tox-cover http://logs.openstack.org/81/639281/5/check/openstack-tox-cover/bca2df0/cover/ : SUCCESS in 17m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/5/check/openstack-tox-lower-constraints/32e1511/ : SUCCESS in 12m 21s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/5/check/openstack-tox-pep8/4ea92b9/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/5/check/openstack-tox-py27/ae7b3f4/ : SUCCESS in 13m 43s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/5/check/openstack-tox-py35/cee10ae/ : SUCCESS in 14m 12s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/5/check/openstack-tox-py36/0bf0de2/ : SUCCESS in 11m 41s\n- openstack-tox-docs http://logs.openstack.org/81/639281/5/check/openstack-tox-docs/3fe4e12/html/ : SUCCESS in 7m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/03e7fbe/ : SUCCESS in 51m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/5/check/devstack-plugin-ceph-tempest/cee4a5e/ : FAILURE in 1h 41m 17s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/5/check/neutron-grenade-multinode/9e35152/ : SUCCESS in 1h 15m 37s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/5/check/neutron-tempest-linuxbridge/e7baf71/ : SUCCESS in 1h 35m 40s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/5/check/nova-grenade-live-migration/8b77d79/ : SUCCESS in 1h 18m 09s\n- nova-live-migration http://logs.openstack.org/81/639281/5/check/nova-live-migration/a42da30/ : SUCCESS in 50m 07s\n- nova-lvm http://logs.openstack.org/81/639281/5/check/nova-lvm/406de54/ : SUCCESS in 1h 04m 03s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/5/check/nova-next/e43345f/ : SUCCESS in 1h 59m 58s\n- nova-tox-functional http://logs.openstack.org/81/639281/5/check/nova-tox-functional/211bc85/ : SUCCESS in 24m 58s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/5/check/nova-tox-functional-py35/f986b03/ : SUCCESS in 21m 13s\n- tempest-multinode-full http://logs.openstack.org/81/639281/5/check/tempest-multinode-full/8f00015/ : SUCCESS in 1h 32m 35s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/5/check/tempest-slow-py3/80e6187/ : SUCCESS in 2h 14m 53s","accounts_in_message":[],"_revision_number":5},{"id":"79732c4f3271a978a185f200bc0bc2d6964e92b4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 09:30:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/5/check/EMC_VxFlexOS_NOVA/48d650b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"9588743c27dc63c3282925e186ae838d285d436c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 14:51:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/5 : SUCCESS in 2h 39m 08s","accounts_in_message":[],"_revision_number":5},{"id":"e696bb31d1dc2a085341de39ba8a2b924c2f04fe","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-03-08 16:19:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/81/639281/5/check-nova/Nova-ML2-Sriov/5d6bddc : SUCCESS in 1h 13m 44s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/81/639281/5/check-nova/Nova-MACVTAP-ML2-Sriov/14a1f89 : SUCCESS in 1h 09m 33s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"efc800b2039cb07918a3063cdd262b7268c962bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-09 01:29:40.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"82f58fed5592c832993cc69571c64e946cb32796","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-09 02:25:54.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/5/check-vote/ext-nova-zuul/1c6c2bc : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":5},{"id":"6497411fe190554a67c7927fcb35905367e1a0ca","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 06:15:32.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/187cb3b : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":5},{"id":"2c56697ef39799b7d7cc392960f8148ee51dc27b","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-18 05:44:49.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8098de2c645ab7bb91902e3731e564aad4ea648e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 05:53:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37026af022d158a8c12434f124f3930e7014b36d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 08:24:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/6/check/grenade-py3/d62ecac/ : SUCCESS in 50m 11s\n- tempest-full-py3 http://logs.openstack.org/81/639281/6/check/tempest-full-py3/8a3547f/ : SUCCESS in 1h 21m 50s\n- openstack-tox-cover http://logs.openstack.org/81/639281/6/check/openstack-tox-cover/ed2b25c/cover/ : SUCCESS in 20m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/6/check/openstack-tox-lower-constraints/0891839/ : SUCCESS in 12m 02s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/6/check/openstack-tox-pep8/9938473/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/6/check/openstack-tox-py27/2b252c0/ : SUCCESS in 12m 29s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/6/check/openstack-tox-py35/687aede/ : SUCCESS in 13m 36s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/6/check/openstack-tox-py36/bf72480/ : SUCCESS in 11m 08s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/6/check/openstack-tox-py37/304fb94/ : SUCCESS in 12m 10s\n- openstack-tox-docs http://logs.openstack.org/81/639281/6/check/openstack-tox-docs/01bd9c2/html/ : SUCCESS in 6m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/01f4c0f/ : SUCCESS in 49m 48s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/6/check/devstack-plugin-ceph-tempest/90c21e4/ : SUCCESS in 1h 37m 45s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/6/check/neutron-grenade-multinode/b7ecb85/ : SUCCESS in 1h 12m 29s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/6/check/neutron-tempest-linuxbridge/4d04fc3/ : SUCCESS in 1h 30m 25s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/6/check/nova-grenade-live-migration/4344c4a/ : SUCCESS in 1h 05m 54s\n- nova-live-migration http://logs.openstack.org/81/639281/6/check/nova-live-migration/7b93a7c/ : SUCCESS in 48m 01s\n- nova-lvm http://logs.openstack.org/81/639281/6/check/nova-lvm/af07f07/ : SUCCESS in 1h 02m 28s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/6/check/nova-next/4427599/ : SUCCESS in 1h 32m 21s\n- nova-tox-functional http://logs.openstack.org/81/639281/6/check/nova-tox-functional/43ba6ef/ : SUCCESS in 20m 16s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/6/check/nova-tox-functional-py35/5bac177/ : SUCCESS in 19m 39s\n- tempest-multinode-full http://logs.openstack.org/81/639281/6/check/tempest-multinode-full/a945d8d/ : SUCCESS in 1h 44m 20s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/6/check/tempest-slow-py3/babbca4/ : SUCCESS in 2h 30m 13s","accounts_in_message":[],"_revision_number":6},{"id":"317a75232af1b99933409feba8033ee92f837ce7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-18 08:41:36.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29286 : SUCCESS in 2h 44m 42s","accounts_in_message":[],"_revision_number":6},{"id":"f785e1252a90301d38c19f09ea0a1d6cbb9e1f91","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 09:04:03.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/6/check-vote/ext-nova-zuul/d0503b5 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":6},{"id":"64f683aa0ff55ebc7f3c6604b942c9bdefd9fcac","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-18 09:58:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/6 : SUCCESS in 2h 19m 11s","accounts_in_message":[],"_revision_number":6},{"id":"99367e3dceac2f5bae7c538ca6b12b75ab117200","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 12:03:13.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2df72f7 : SUCCESS in 1h 33m 58s","accounts_in_message":[],"_revision_number":6},{"id":"5c98eda257a7d2a5a9416baf8955d38aecee9b62","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-18 14:33:58.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"89141bb76c3c521f0c40b25d72ecf150c7300709","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-18 19:26:35.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/6/check/tempest-dsvm-full-xenial/296d406/ : SUCCESS in 1h 19m 02s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/6/check/tempest-dsvm-full-xenial-py3/4c46642/ : SUCCESS in 1h 17m 58s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/6/check/grenade-dsvm-xenial/f8ef065/ : SUCCESS in 1h 02m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0c2b39ebb7f091f6747e220be40bba97d5169355","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-21 09:37:18.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"a10f65f6c2ae8bae9295fd569a734aaf8abb3432","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 09:51:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"710cba4f0c55bcc74c21bf0314a3026a4742d21e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:24:18.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/7/check/nova-out-of-tree-pvm/56f8146 : FAILURE in 39m 47s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/7/check/nova-in-tree-pvm/6b524c7 : FAILURE in 39m 17s","accounts_in_message":[],"_revision_number":7},{"id":"64a25a08eb8162aaea39d83f39eaa10175a64abd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-21 11:23:10.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/7/check/tempest-dsvm-full-xenial/7a59ab6/ : SUCCESS in 1h 14m 09s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/7/check/tempest-dsvm-full-xenial-py3/825cd88/ : SUCCESS in 1h 21m 32s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/7/check/grenade-dsvm-xenial/e41f4c9/ : SUCCESS in 58m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"727e09a1f22227e21bec111f885cec9e3f922cc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 11:50:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/7/check/grenade-py3/07dcf50/ : SUCCESS in 1h 01m 48s\n- tempest-full-py3 http://logs.openstack.org/81/639281/7/check/tempest-full-py3/6df9f24/ : SUCCESS in 1h 48m 48s\n- openstack-tox-cover http://logs.openstack.org/81/639281/7/check/openstack-tox-cover/cd67a1d/cover/ : SUCCESS in 29m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/7/check/openstack-tox-lower-constraints/e90607b/ : SUCCESS in 14m 54s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/7/check/openstack-tox-pep8/988cf38/ : SUCCESS in 11m 09s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/7/check/openstack-tox-py27/04fef93/ : SUCCESS in 15m 45s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/7/check/openstack-tox-py35/e93b473/ : SUCCESS in 14m 12s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/7/check/openstack-tox-py36/2ecd952/ : SUCCESS in 15m 08s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/7/check/openstack-tox-py37/8129841/ : SUCCESS in 18m 23s\n- openstack-tox-docs http://logs.openstack.org/81/639281/7/check/openstack-tox-docs/f7e64df/html/ : SUCCESS in 8m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/3b38d06/ : SUCCESS in 49m 24s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/7/check/devstack-plugin-ceph-tempest/5464a8b/ : SUCCESS in 1h 38m 44s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/7/check/neutron-grenade-multinode/f2cdea6/ : SUCCESS in 1h 26m 51s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/7/check/neutron-tempest-linuxbridge/e49c46d/ : SUCCESS in 1h 20m 52s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/7/check/nova-grenade-live-migration/b73c2cc/ : SUCCESS in 1h 19m 35s\n- nova-live-migration http://logs.openstack.org/81/639281/7/check/nova-live-migration/ac754ab/ : SUCCESS in 43m 55s\n- nova-lvm http://logs.openstack.org/81/639281/7/check/nova-lvm/43cd449/ : SUCCESS in 51m 49s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/7/check/nova-next/99c874a/ : SUCCESS in 1h 17m 51s\n- nova-tox-functional http://logs.openstack.org/81/639281/7/check/nova-tox-functional/2be3153/ : SUCCESS in 19m 34s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/7/check/nova-tox-functional-py35/3d371dc/ : SUCCESS in 20m 58s\n- tempest-multinode-full http://logs.openstack.org/81/639281/7/check/tempest-multinode-full/23d12c8/ : SUCCESS in 1h 39m 03s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639281/7/check/tempest-slow-py3/983f1aa/ : SUCCESS in 2h 05m 06s","accounts_in_message":[],"_revision_number":7},{"id":"bfff4f4eb9bd024b8d8dd0de7b2be20ecc182c31","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 11:52:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/7/check/EMC_VxFlexOS_NOVA/1cc8fa6/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"db3d98aa32a3da6fce040823bda70bedf1354720","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-21 12:09:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/7 : SUCCESS in 2h 17m 29s","accounts_in_message":[],"_revision_number":7},{"id":"046125c592b0c65550a7218b5dc7d253dba9e113","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-21 13:22:45.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29429 : SUCCESS in 3h 28m 38s","accounts_in_message":[],"_revision_number":7},{"id":"bfd694515b86c7c263abd5d84763c7a1f117dae3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 16:58:57.000000000","message":"Patch Set 7:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/f46facc : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":7},{"id":"24447094515d86b5f1d3438c19f98650be9329fc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-21 17:11:38.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"bb09c3e182b93594209eeef926a1f3e8d6f7b289","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-21 17:46:46.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"df0c28f870140218fcdbe44f50a8a5dda8cde6b7","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-27 04:38:51.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"317bbda76397b81944b55cc9e61278bfcbac6129","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 04:47:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d3a39ac7b13f5d70c9299435d1c34f7d43480c21","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-27 06:02:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/8/check/tempest-dsvm-full-xenial/f9ab4d5/ : SUCCESS in 1h 10m 59s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/8/check/tempest-dsvm-full-xenial-py3/7fa53ec/ : SUCCESS in 1h 12m 31s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/8/check/grenade-dsvm-xenial/b5bf81c/ : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9acf7400faf911aea06987666a6fb8b430d58706","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:10:07.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/8/check/nova-out-of-tree-pvm/0fc6ad0 : SUCCESS in 2h 11m 01s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/8/check/nova-in-tree-pvm/9d38e9d : FAILURE in 2h 25m 54s","accounts_in_message":[],"_revision_number":8},{"id":"8b47ddbfc72eb11d5e7d648246e1935dc1fd12b9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-27 07:11:15.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/8 : FAILURE in 2h 23m 49s","accounts_in_message":[],"_revision_number":8},{"id":"9f9eb03a6860a731a6e29419481a6b95492e2a37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 07:11:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/8/check/grenade-py3/1b3204d/ : SUCCESS in 58m 18s\n- tempest-full-py3 http://logs.openstack.org/81/639281/8/check/tempest-full-py3/aefd82e/ : SUCCESS in 1h 33m 57s\n- openstack-tox-cover http://logs.openstack.org/81/639281/8/check/openstack-tox-cover/38e349a/cover/ : SUCCESS in 17m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/8/check/openstack-tox-lower-constraints/39ea4aa/ : SUCCESS in 13m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/8/check/openstack-tox-pep8/66863e9/ : SUCCESS in 10m 29s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/8/check/openstack-tox-py27/81f40d4/ : SUCCESS in 14m 43s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/8/check/openstack-tox-py35/b7914ea/ : SUCCESS in 12m 54s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/8/check/openstack-tox-py36/de7f7a6/ : SUCCESS in 12m 52s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/8/check/openstack-tox-py37/314aea1/ : SUCCESS in 13m 12s\n- openstack-tox-docs http://logs.openstack.org/81/639281/8/check/openstack-tox-docs/c223500/html/ : SUCCESS in 8m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8e155cd/ : SUCCESS in 50m 50s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/8/check/devstack-plugin-ceph-tempest/4b468a9/ : SUCCESS in 1h 31m 48s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/8/check/neutron-grenade-multinode/9f73122/ : SUCCESS in 1h 15m 57s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/8/check/neutron-tempest-linuxbridge/31168cc/ : SUCCESS in 1h 31m 31s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/8/check/nova-grenade-live-migration/78917f2/ : SUCCESS in 1h 11m 23s\n- nova-live-migration http://logs.openstack.org/81/639281/8/check/nova-live-migration/2a32454/ : SUCCESS in 47m 40s\n- nova-lvm http://logs.openstack.org/81/639281/8/check/nova-lvm/60333b9/ : SUCCESS in 1h 04m 20s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/8/check/nova-next/ed37532/ : SUCCESS in 1h 17m 26s\n- nova-tox-functional http://logs.openstack.org/81/639281/8/check/nova-tox-functional/5c74bab/ : SUCCESS in 19m 34s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/8/check/nova-tox-functional-py35/dffe029/ : SUCCESS in 21m 51s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/8/check/tempest-slow-py3/214ecf6/ : SUCCESS in 2h 12m 32s","accounts_in_message":[],"_revision_number":8},{"id":"8798d0dd68f249e1a8d44c0468d2d0b80a78a9ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:13:17.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/8/check/EMC_VxFlexOS_NOVA/6ba8745/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"09b37a05135dff3ace3e6ffc828c5b9270fb92a7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-27 07:26:57.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29581 : SUCCESS in 2h 36m 14s","accounts_in_message":[],"_revision_number":8},{"id":"aecbeebe9a1962fda23b1c65bdbc3cf67b10ad47","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 08:54:56.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5c5f66b : FAILURE in 26m 05s","accounts_in_message":[],"_revision_number":8},{"id":"e11e0b3020a50bc459fb2e2accfbd4dc4bfd626a","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-03-27 09:47:00.000000000","message":"Patch Set 8:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":8},{"id":"7c9149358a0f64eef2c8f4ac521f2b339eb508fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-27 11:03:20.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"2b2cfa3ed25cc52eaba925fbee9599178fbc5ec1","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 04:13:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"2a9b59ea12cfbe5fe85732c1389894422d84f52c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 04:29:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0f2dd00a3566daa983438df4d4448d7d0ea2c4ef","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 05:55:58.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/9/check/tempest-dsvm-full-xenial/1d5e92e/ : SUCCESS in 1h 14m 54s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/9/check/tempest-dsvm-full-xenial-py3/23ea912/ : SUCCESS in 1h 15m 01s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/9/check/grenade-dsvm-xenial/31930da/ : SUCCESS in 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cca548c9ea15f56ea46aa01cb41b50a3fd3b568d","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 06:09:26.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"8d7b59494d67169f92d16c6be544670bf510cf72","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 06:30:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ea198059b5a5219a37ed7a384eaaa391e7fc122e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 08:06:51.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/10/check/tempest-dsvm-full-xenial/721c902/ : SUCCESS in 1h 08m 40s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/10/check/tempest-dsvm-full-xenial-py3/ed6fd0b/ : SUCCESS in 1h 21m 08s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/10/check/grenade-dsvm-xenial/a756bca/ : SUCCESS in 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"507d3508b4acda7787ef808b4229695bd48cd532","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 08:40:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/10/check/grenade-py3/2a9c335/ : SUCCESS in 54m 04s\n- tempest-full-py3 http://logs.openstack.org/81/639281/10/check/tempest-full-py3/cbaa2f2/ : SUCCESS in 1h 39m 52s\n- openstack-tox-cover http://logs.openstack.org/81/639281/10/check/openstack-tox-cover/9366cf2/cover/ : SUCCESS in 16m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/10/check/openstack-tox-lower-constraints/a268fbf/ : SUCCESS in 14m 05s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/10/check/openstack-tox-pep8/8270ac0/ : FAILURE in 10m 21s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/10/check/openstack-tox-py27/9a19778/ : SUCCESS in 13m 42s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/10/check/openstack-tox-py35/c53af2e/ : SUCCESS in 13m 45s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/10/check/openstack-tox-py36/945da92/ : SUCCESS in 12m 13s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/10/check/openstack-tox-py37/4a34aec/ : SUCCESS in 13m 34s\n- openstack-tox-docs http://logs.openstack.org/81/639281/10/check/openstack-tox-docs/c6b6276/html/ : SUCCESS in 6m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/10d6da4/ : SUCCESS in 49m 31s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/10/check/devstack-plugin-ceph-tempest/890af6f/ : SUCCESS in 1h 33m 50s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/10/check/neutron-grenade-multinode/43e5f83/ : SUCCESS in 1h 08m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/10/check/neutron-tempest-linuxbridge/c133b26/ : SUCCESS in 1h 23m 15s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/10/check/nova-grenade-live-migration/13a4dcd/ : SUCCESS in 1h 06m 16s\n- nova-live-migration http://logs.openstack.org/81/639281/10/check/nova-live-migration/04fc825/ : SUCCESS in 46m 47s\n- nova-lvm http://logs.openstack.org/81/639281/10/check/nova-lvm/defb570/ : SUCCESS in 52m 14s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/10/check/nova-next/77dfdc9/ : SUCCESS in 1h 10m 29s\n- nova-tox-functional http://logs.openstack.org/81/639281/10/check/nova-tox-functional/702def0/ : SUCCESS in 18m 23s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/10/check/nova-tox-functional-py35/e8bfa9c/ : SUCCESS in 18m 26s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/10/check/tempest-slow-py3/2c0d744/ : SUCCESS in 2h 06m 19s","accounts_in_message":[],"_revision_number":10},{"id":"d1329a64b315d16e476d5b6af66d0add5b263b60","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 09:24:50.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29678 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":10},{"id":"bd9cceb8fc6ebf60975270e3382ea57b3ab4308c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-29 11:07:46.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/10/check/nova-out-of-tree-pvm/b79b219 : FAILURE in 2h 57m 38s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/10/check/nova-in-tree-pvm/5200a76 : FAILURE in 3h 09m 33s","accounts_in_message":[],"_revision_number":10},{"id":"6d0c29bebd965bf9e2d79aa554a97f3811861316","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 11:33:05.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/10 : FAILURE in 2h 48m 37s","accounts_in_message":[],"_revision_number":10},{"id":"4d05c12d4f49d03f61a600db4839503581696e93","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 12:14:56.000000000","message":"Patch Set 10:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/abf1b5b : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":10},{"id":"ac3339559a7a5fd63c73978d5ededb5bbdbbf3b3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 20:45:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"ee93ceb087818bf5bc93ea24ad18d4b701b819c5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 21:03:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f4ad7baa16e333032ff857037c619213c6acb44e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 22:34:24.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/11/check/tempest-dsvm-full-xenial/a9da284/ : SUCCESS in 1h 18m 08s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/11/check/tempest-dsvm-full-xenial-py3/d39e8c0/ : SUCCESS in 1h 14m 21s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/11/check/grenade-dsvm-xenial/040143f/ : SUCCESS in 57m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ad867131ec2aea6256ec4c4b8fe54d59158690e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 00:07:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/11/check/grenade-py3/a0bbcf7/ : SUCCESS in 1h 01m 26s\n- tempest-full-py3 http://logs.openstack.org/81/639281/11/check/tempest-full-py3/0aba6ee/ : SUCCESS in 1h 40m 00s\n- openstack-tox-cover http://logs.openstack.org/81/639281/11/check/openstack-tox-cover/f5ee41e/cover/ : SUCCESS in 14m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/11/check/openstack-tox-lower-constraints/14b4bd8/ : SUCCESS in 13m 04s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/11/check/openstack-tox-pep8/27ab20a/ : SUCCESS in 11m 03s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/11/check/openstack-tox-py27/19daf02/ : SUCCESS in 13m 06s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/11/check/openstack-tox-py35/7feaaff/ : SUCCESS in 13m 11s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/11/check/openstack-tox-py36/0a9ecff/ : SUCCESS in 12m 59s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/11/check/openstack-tox-py37/80b79da/ : SUCCESS in 12m 19s\n- openstack-tox-docs http://logs.openstack.org/81/639281/11/check/openstack-tox-docs/2a85ca0/html/ : SUCCESS in 6m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/fe19455/ : SUCCESS in 58m 30s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/11/check/devstack-plugin-ceph-tempest/6c3faeb/ : SUCCESS in 1h 34m 05s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/11/check/neutron-grenade-multinode/144009c/ : SUCCESS in 1h 09m 53s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/11/check/neutron-tempest-linuxbridge/0c922bf/ : SUCCESS in 1h 32m 22s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/11/check/nova-grenade-live-migration/4805223/ : SUCCESS in 1h 02m 04s\n- nova-live-migration http://logs.openstack.org/81/639281/11/check/nova-live-migration/0bb80e6/ : SUCCESS in 53m 30s\n- nova-lvm http://logs.openstack.org/81/639281/11/check/nova-lvm/174cdf3/ : SUCCESS in 1h 08m 03s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/11/check/nova-next/35cd271/ : SUCCESS in 1h 26m 54s\n- nova-tox-functional http://logs.openstack.org/81/639281/11/check/nova-tox-functional/7cc7663/ : SUCCESS in 21m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/11/check/nova-tox-functional-py35/83dfcd6/ : SUCCESS in 20m 02s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/11/check/tempest-slow-py3/8c5dce9/ : SUCCESS in 2h 21m 48s","accounts_in_message":[],"_revision_number":11},{"id":"0b21ed12b07331ca1f73adaa810b84e13b11047d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 00:08:50.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/11/check/EMC_VxFlexOS_NOVA/e1f8d27/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"986582f0497d74378b9a068859b6272a1bb28dcf","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-30 00:27:08.000000000","message":"Patch Set 11:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/11/check/nova-out-of-tree-pvm/096d0d3 : SUCCESS in 2h 39m 12s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/11/check/nova-in-tree-pvm/3b785b1 : FAILURE in 2h 27m 39s","accounts_in_message":[],"_revision_number":11},{"id":"e0eaa7f54d94ea5fe65557d6be1b8f52a76a969e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-30 00:32:10.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29720 : SUCCESS in 3h 23m 20s","accounts_in_message":[],"_revision_number":11},{"id":"270877b205582df0dac8f4a2fd1d43ef65af5c26","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-30 01:13:21.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/11 : FAILURE in 2h 08m 07s","accounts_in_message":[],"_revision_number":11},{"id":"e4622656a71e86962b4e30ec7452ceed404d320b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-30 03:13:46.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"7e666a6c5cadbd1035971aa9df0a97066a16bcb3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 15:04:21.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/3af1df3 : SUCCESS in 1h 40m 01s","accounts_in_message":[],"_revision_number":11},{"id":"6b86e316772f3edd1e2cd12d41d7ae54aea5bd9f","author":{"_account_id":30103,"name":"xiaosa","email":"xiaosa@awcloud.com","username":"xiaosa"},"date":"2019-04-01 06:44:59.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"b26522922f711aa86693055d7fa9d54bc6b7c4d5","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-01 21:18:06.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"cd14c168b5ea947d266fd7503847c7b34a505e3b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 21:41:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6ca48652eff0fa5fcc9d36fe370e69784268926f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-01 23:19:07.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/12/check/tempest-dsvm-full-xenial/c09c702/ : SUCCESS in 1h 11m 33s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/12/check/tempest-dsvm-full-xenial-py3/806247d/ : SUCCESS in 1h 17m 53s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/12/check/grenade-dsvm-xenial/ce6d954/ : SUCCESS in 59m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5d2d51be416cbb4beee98a5d1bcffbd09f28755a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 23:57:08.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/12/check/grenade-py3/c5c3e33/ : SUCCESS in 1h 29m 11s\n- tempest-full-py3 http://logs.openstack.org/81/639281/12/check/tempest-full-py3/0619641/ : SUCCESS in 1h 46m 01s\n- openstack-tox-cover http://logs.openstack.org/81/639281/12/check/openstack-tox-cover/0fbd12e/ : FAILURE in 21m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/12/check/openstack-tox-lower-constraints/302c95b/ : FAILURE in 15m 26s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/12/check/openstack-tox-pep8/b211d96/ : SUCCESS in 12m 55s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/12/check/openstack-tox-py27/f6182c4/ : SUCCESS in 12m 17s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/12/check/openstack-tox-py35/7cf69de/ : FAILURE in 13m 35s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/12/check/openstack-tox-py36/310db9c/ : FAILURE in 12m 43s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/12/check/openstack-tox-py37/bb12622/ : FAILURE in 12m 03s\n- openstack-tox-docs http://logs.openstack.org/81/639281/12/check/openstack-tox-docs/03d9c45/html/ : SUCCESS in 6m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a261214/ : SUCCESS in 49m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/12/check/devstack-plugin-ceph-tempest/1d488da/ : SUCCESS in 1h 46m 13s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/12/check/neutron-grenade-multinode/66ae142/ : SUCCESS in 1h 19m 46s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/12/check/neutron-tempest-linuxbridge/88ad4d4/ : SUCCESS in 1h 26m 32s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/12/check/nova-grenade-live-migration/0d8e27a/ : SUCCESS in 1h 08m 45s\n- nova-live-migration http://logs.openstack.org/81/639281/12/check/nova-live-migration/b8881f3/ : FAILURE in 50m 21s\n- nova-lvm http://logs.openstack.org/81/639281/12/check/nova-lvm/13e796d/ : SUCCESS in 1h 00m 55s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/12/check/nova-next/427d82f/ : SUCCESS in 1h 44m 53s\n- nova-tox-functional http://logs.openstack.org/81/639281/12/check/nova-tox-functional/ebd27af/ : SUCCESS in 21m 08s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/12/check/nova-tox-functional-py35/a14293a/ : SUCCESS in 18m 27s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/12/check/tempest-slow-py3/c08fd60/ : SUCCESS in 2h 04m 42s","accounts_in_message":[],"_revision_number":12},{"id":"473377b5a2925efa3266c86cbca618e7e12cd5d5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 00:49:47.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29795 : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":12},{"id":"4fbc26364303243c782cdd591a0e92378e660fc2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 02:30:50.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/12 : SUCCESS in 2h 39m 04s","accounts_in_message":[],"_revision_number":12},{"id":"01a20aa114b140aa59ae4d0c7b66a4a555174fcc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 02:31:41.000000000","message":"Patch Set 12:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/36bf303 : FAILURE in 20m 47s","accounts_in_message":[],"_revision_number":12},{"id":"826812b3c3ede872d03f8708458482d330ad7053","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-02 10:13:10.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"9428eddef437e69174cd726dc5fcef4062aead70","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 10:41:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cb3259416c7b68e4ffd8c6dc0da0bb75931b0118","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 10:46:07.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/13 : FAILURE in 45s","accounts_in_message":[],"_revision_number":13},{"id":"8445b3fc6eb0b72c235897edb1f09f7b1fef365d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 13:26:30.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/13/check/grenade-py3/d930274/ : SUCCESS in 1h 01m 38s\n- tempest-full-py3 http://logs.openstack.org/81/639281/13/check/tempest-full-py3/2302981/ : SUCCESS in 1h 52m 33s\n- openstack-tox-cover http://logs.openstack.org/81/639281/13/check/openstack-tox-cover/2824fda/ : FAILURE in 13m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/13/check/openstack-tox-lower-constraints/c9671ba/ : FAILURE in 12m 15s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/13/check/openstack-tox-pep8/6fec214/ : SUCCESS in 9m 52s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/13/check/openstack-tox-py27/d45a5d7/ : SUCCESS in 12m 41s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/13/check/openstack-tox-py35/a36db07/ : FAILURE in 12m 35s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/13/check/openstack-tox-py36/898aa62/ : FAILURE in 12m 14s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/13/check/openstack-tox-py37/e6aa8a0/ : FAILURE in 14m 09s\n- openstack-tox-docs http://logs.openstack.org/81/639281/13/check/openstack-tox-docs/1961af2/html/ : SUCCESS in 7m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/13/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d7bf465/ : SUCCESS in 49m 16s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/13/check/devstack-plugin-ceph-tempest/f6aec85/ : SUCCESS in 1h 31m 02s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/13/check/neutron-grenade-multinode/8b43ce7/ : SUCCESS in 1h 04m 51s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/13/check/neutron-tempest-linuxbridge/5c5a8a4/ : SUCCESS in 1h 26m 36s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/13/check/nova-grenade-live-migration/36a4660/ : SUCCESS in 1h 03m 18s\n- nova-live-migration http://logs.openstack.org/81/639281/13/check/nova-live-migration/8b4c5c5/ : SUCCESS in 1h 01m 10s\n- nova-lvm http://logs.openstack.org/81/639281/13/check/nova-lvm/37cea28/ : SUCCESS in 1h 09m 47s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/13/check/nova-next/6b627bf/ : SUCCESS in 1h 34m 56s\n- nova-tox-functional http://logs.openstack.org/81/639281/13/check/nova-tox-functional/b8dc0b7/ : SUCCESS in 21m 42s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/13/check/nova-tox-functional-py35/d0726af/ : SUCCESS in 21m 16s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/13/check/tempest-slow-py3/1a89cc0/ : SUCCESS in 2h 14m 32s","accounts_in_message":[],"_revision_number":13},{"id":"8358d2f2d26606fa554fc838b19ecfcddb044dc8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 14:23:43.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29884 : FAILURE in 23m 05s","accounts_in_message":[],"_revision_number":13},{"id":"ccf74ec2d4e2b19d3d3ef6e4f74d29535d34006b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-02 20:17:44.000000000","message":"Patch Set 13:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/13/check/tempest-dsvm-full-xenial/07ee694/ : SUCCESS in 1h 15m 23s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/13/check/tempest-dsvm-full-xenial-py3/e74e445/ : SUCCESS in 1h 26m 23s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/13/check/grenade-dsvm-xenial/f6bfeb2/ : SUCCESS in 1h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c9e089f56cfab01e27f0f1a46937061fc5269787","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-03 08:24:07.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/40c964e : SUCCESS in 1h 41m 47s","accounts_in_message":[],"_revision_number":13},{"id":"5aa36529cfda688681c5ccbc047bb02e7c26a2fb","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 08:44:44.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"0245496b860e6cdf77f2eb53f9444fc5c64725b8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-04 09:09:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ab2c53d1d8ffe53dcc6f01f6859f2876086d1cd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 12:02:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/14/check/grenade-py3/e6a2a26/ : SUCCESS in 1h 02m 31s\n- tempest-full-py3 http://logs.openstack.org/81/639281/14/check/tempest-full-py3/bba153a/ : SUCCESS in 1h 40m 12s\n- openstack-tox-cover http://logs.openstack.org/81/639281/14/check/openstack-tox-cover/a17bc24/ : FAILURE in 19m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/14/check/openstack-tox-lower-constraints/77054ab/ : FAILURE in 17m 14s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/14/check/openstack-tox-pep8/3f4c5fd/ : SUCCESS in 12m 22s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/14/check/openstack-tox-py27/df56807/ : SUCCESS in 17m 28s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/14/check/openstack-tox-py35/af848c9/ : FAILURE in 16m 41s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/14/check/openstack-tox-py36/e261cf4/ : FAILURE in 13m 02s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/14/check/openstack-tox-py37/5d99b42/ : FAILURE in 17m 49s\n- openstack-tox-docs http://logs.openstack.org/81/639281/14/check/openstack-tox-docs/1f2a1e6/html/ : SUCCESS in 6m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/14/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6193fad/ : SUCCESS in 45m 00s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/14/check/devstack-plugin-ceph-tempest/37090a3/ : SUCCESS in 1h 41m 54s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/14/check/neutron-grenade-multinode/2caac79/ : SUCCESS in 1h 00m 06s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/14/check/neutron-tempest-linuxbridge/38a0797/ : SUCCESS in 1h 47m 42s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/14/check/nova-grenade-live-migration/9e2596d/ : SUCCESS in 1h 06m 09s\n- nova-live-migration http://logs.openstack.org/81/639281/14/check/nova-live-migration/21d196f/ : SUCCESS in 1h 57m 04s\n- nova-lvm http://logs.openstack.org/81/639281/14/check/nova-lvm/7c0d896/ : SUCCESS in 1h 00m 17s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/14/check/nova-next/046faed/ : SUCCESS in 1h 25m 31s\n- nova-tox-functional http://logs.openstack.org/81/639281/14/check/nova-tox-functional/ff0d6d1/ : SUCCESS in 19m 53s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/14/check/nova-tox-functional-py35/65a15c6/ : SUCCESS in 22m 12s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/14/check/tempest-slow-py3/45a267d/ : SUCCESS in 2h 16m 36s","accounts_in_message":[],"_revision_number":14},{"id":"2e860006d8303d54e46b3653aa4254ffe8b1e552","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-04 18:17:32.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30032 : SUCCESS in 3h 27m 42s","accounts_in_message":[],"_revision_number":14},{"id":"dc4382153bd06f7c5b47209f058a30e212e7d3f2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 18:18:43.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/14/check/tempest-dsvm-full-xenial/71a4925/ : SUCCESS in 1h 28m 31s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/14/check/tempest-dsvm-full-xenial-py3/f12e780/ : SUCCESS in 1h 18m 41s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/14/check/grenade-dsvm-xenial/38504f7/ : SUCCESS in 1h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"475bdb3f2ad06c25169dd5d398e40a8eb5cd2a16","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 20:44:48.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"d57b466505a091ee6656d25b9213a15bcd1f51e0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 22:53:12.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/15/check/tempest-dsvm-full-xenial/a30c9ca/ : SUCCESS in 1h 21m 45s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/15/check/tempest-dsvm-full-xenial-py3/ddcf52f/ : SUCCESS in 1h 16m 43s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/15/check/grenade-dsvm-xenial/84b7c2c/ : SUCCESS in 1h 00m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"055d32ec015457428b421d42dd168dfcab9af460","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-04 23:05:19.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30073 : FAILURE in 1h 19m 16s","accounts_in_message":[],"_revision_number":15},{"id":"8250df950ff882c846091c8624152eb366b3469d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 23:56:30.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/15/check/grenade-py3/6c483f8/ : SUCCESS in 51m 16s\n- tempest-full-py3 http://logs.openstack.org/81/639281/15/check/tempest-full-py3/f7bfc77/ : SUCCESS in 1h 31m 58s\n- openstack-tox-cover http://logs.openstack.org/81/639281/15/check/openstack-tox-cover/9586a52/cover/ : SUCCESS in 17m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/15/check/openstack-tox-lower-constraints/5176b09/ : SUCCESS in 16m 16s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/15/check/openstack-tox-pep8/f4c4c94/ : SUCCESS in 10m 57s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/15/check/openstack-tox-py27/5ec2e5c/ : SUCCESS in 16m 44s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/15/check/openstack-tox-py35/453a1ed/ : SUCCESS in 15m 49s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/15/check/openstack-tox-py36/80f9b47/ : SUCCESS in 12m 33s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/15/check/openstack-tox-py37/2a020e3/ : SUCCESS in 13m 53s\n- openstack-tox-docs http://logs.openstack.org/81/639281/15/check/openstack-tox-docs/3bcec1e/html/ : SUCCESS in 8m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/15/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/16d3b75/ : SUCCESS in 54m 58s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/15/check/devstack-plugin-ceph-tempest/dacb8f8/ : SUCCESS in 1h 31m 01s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/15/check/neutron-grenade-multinode/2bcfc2e/ : FAILURE in 36m 42s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/15/check/neutron-tempest-linuxbridge/8964b75/ : SUCCESS in 1h 33m 26s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/15/check/nova-grenade-live-migration/3963123/ : SUCCESS in 1h 05m 41s\n- nova-live-migration http://logs.openstack.org/81/639281/15/check/nova-live-migration/03c663c/ : SUCCESS in 51m 02s\n- nova-lvm http://logs.openstack.org/81/639281/15/check/nova-lvm/addf5c0/ : SUCCESS in 1h 11m 17s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/15/check/nova-next/d523761/ : SUCCESS in 1h 45m 54s\n- nova-tox-functional http://logs.openstack.org/81/639281/15/check/nova-tox-functional/6176676/ : SUCCESS in 22m 38s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/15/check/nova-tox-functional-py35/98f515b/ : SUCCESS in 22m 32s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/15/check/tempest-slow-py3/9040203/ : SUCCESS in 2h 19m 07s","accounts_in_message":[],"_revision_number":15},{"id":"400d871213ef0e91d8e328185d1705157cff6ba4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 00:02:35.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/15/check/EMC_VxFlexOS_NOVA/8c6514e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"be260c19d02ad00416eba5fc341d72c3061fa115","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-05 02:17:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/15 : SUCCESS in 2h 30m 40s","accounts_in_message":[],"_revision_number":15},{"id":"b4228596ae51a5ea9cd0b1a67b062ac2c616049e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-05 07:12:28.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"0d7e26a15a9627d6edb4e828e880d754c653b41d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-05 14:52:39.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/15/check-vote/ext-nova-zuul/6d3111a : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":15},{"id":"8f041f5df7705a76849f72e6b1eabaa927ca3f9f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-07 04:00:35.000000000","message":"Patch Set 15:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/02cccb9 : FAILURE in 20m 44s","accounts_in_message":[],"_revision_number":15},{"id":"de99055041a309b88982d5bdf25f2101d80afc8e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-07 23:58:19.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/15/check/check-dsvm-tempest-vz7-exe-minimal/5082b74 : FAILURE in 1h 07m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"404706e127909fde2330a663ae4c57a6d14279d0","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-12 06:57:52.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"43d627e345209d46e2f603979a6f6693e40e503b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 07:10:15.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"2e98483072cfeedc3679f767010f19f02c6f29e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 10:07:52.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/16/check/grenade-py3/3bee55a/ : SUCCESS in 57m 52s\n- tempest-full-py3 http://logs.openstack.org/81/639281/16/check/tempest-full-py3/a9771c8/ : SUCCESS in 1h 36m 15s\n- openstack-tox-cover http://logs.openstack.org/81/639281/16/check/openstack-tox-cover/b557cb6/cover/ : SUCCESS in 22m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/16/check/openstack-tox-lower-constraints/78f11b8/ : SUCCESS in 20m 22s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/16/check/openstack-tox-pep8/f7e4fa1/ : SUCCESS in 13m 09s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/16/check/openstack-tox-py27/b0c2302/ : SUCCESS in 16m 41s\n- openstack-tox-py35 http://logs.openstack.org/81/639281/16/check/openstack-tox-py35/29d1b80/ : SUCCESS in 15m 32s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/16/check/openstack-tox-py36/568b4b3/ : SUCCESS in 12m 50s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/16/check/openstack-tox-py37/add5a6a/ : SUCCESS in 13m 39s\n- openstack-tox-docs http://logs.openstack.org/81/639281/16/check/openstack-tox-docs/4f06dd5/html/ : SUCCESS in 8m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/16/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a19d4ed/ : SUCCESS in 52m 45s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/16/check/devstack-plugin-ceph-tempest/d5ab58d/ : SUCCESS in 1h 28m 33s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/16/check/neutron-grenade-multinode/4e3ff28/ : SUCCESS in 1h 08m 10s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/16/check/neutron-tempest-linuxbridge/6ae3f76/ : SUCCESS in 1h 42m 53s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/16/check/nova-grenade-live-migration/07a9069/ : SUCCESS in 1h 09m 05s\n- nova-live-migration http://logs.openstack.org/81/639281/16/check/nova-live-migration/b1c335e/ : SUCCESS in 54m 48s\n- nova-lvm http://logs.openstack.org/81/639281/16/check/nova-lvm/5904bec/ : SUCCESS in 55m 50s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/16/check/nova-next/829392f/ : SUCCESS in 1h 36m 30s\n- nova-tox-functional http://logs.openstack.org/81/639281/16/check/nova-tox-functional/4a03d1a/ : SUCCESS in 26m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639281/16/check/nova-tox-functional-py35/628a7e1/ : FAILURE in 26m 18s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/16/check/tempest-slow-py3/7a4fb96/ : SUCCESS in 2h 21m 06s","accounts_in_message":[],"_revision_number":16},{"id":"628f144f0ba6e23ce1e157396297b4e4301afa45","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-12 10:37:32.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639281/16 : FAILURE in 17m 45s","accounts_in_message":[],"_revision_number":16},{"id":"cd3be9571eedeb236ba26b939609f6bc1b12c5b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 13:25:25.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"9592e08cfed7a5e2d6a5b05cc2c075d54d03a970","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-12 15:14:41.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30431 : SUCCESS in 1h 30m 51s","accounts_in_message":[],"_revision_number":16},{"id":"f2be24e90b93457ba05049d30d93ca24da96b5fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-12 21:47:17.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/16/check/tempest-dsvm-full-xenial/1d7e473/ : SUCCESS in 1h 23m 41s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/16/check/tempest-dsvm-full-xenial-py3/576a8fa/ : SUCCESS in 2h 01m 52s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/16/check/grenade-dsvm-xenial/02aa5c3/ : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6539f02c974489afdf82edb298eacfe2700b1113","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-16 15:59:53.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/16/check/check-dsvm-tempest-vz7-exe-minimal/b068b7c : SUCCESS in 1h 39m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"0d9563160386cf337026305c5fcdd2d949ca7aa0","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-24 01:44:56.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"2b7f596bbc266ecefd28854f5225fa13f8099948","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 01:55:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9ca3308ceaeec36bc333095c3b96ebcd7975280e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:03:12.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/17/check/check-dsvm-tempest-vz7-exe-minimal/39f37d0 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"8d676479854862ab4fb0ce5e2174f34445a69306","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 02:06:20.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/17/check/tempest-dsvm-full-xenial/bbf9aa0/ : FAILURE in 11s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/17/check/tempest-dsvm-full-xenial-py3/48ee392/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/17/check/grenade-dsvm-xenial/ea8b7f1/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"cd759b5f561cc65667723502a71678fc649b05a5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 02:12:31.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30971 : FAILURE in 7s","accounts_in_message":[],"_revision_number":17},{"id":"4c36240b6fdec5415289fc7a2af246db00caa7d2","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:15:58.000000000","message":"Patch Set 17:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e7fafc8 : FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":17},{"id":"84fe41296a9ba75c7d43a0521194014b2e0c25b4","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-24 05:33:50.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/17 : SUCCESS in 2h 12m 15s","accounts_in_message":[],"_revision_number":17},{"id":"071ae77df383b7ca4c5f4d7d750e6be094626b1a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-24 05:42:48.000000000","message":"Patch Set 17:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/17/check/nova-out-of-tree-pvm/2b6b4f1 : FAILURE in 1h 54m 25s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/17/check/nova-in-tree-pvm/f540fee : FAILURE in 1m 17s","accounts_in_message":[],"_revision_number":17},{"id":"e5f5c0261346774ea3c5f717b42720156875ec29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 06:58:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/17/check/grenade-py3/09f1c15/ : SUCCESS in 49m 56s\n- tempest-full-py3 http://logs.openstack.org/81/639281/17/check/tempest-full-py3/df863e4/ : SUCCESS in 1h 34m 57s\n- openstack-tox-cover http://logs.openstack.org/81/639281/17/check/openstack-tox-cover/2ee622c/cover/ : SUCCESS in 14m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/17/check/openstack-tox-lower-constraints/4b4d03c/ : FAILURE in 15m 32s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/17/check/openstack-tox-pep8/7200fca/ : SUCCESS in 10m 06s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/17/check/openstack-tox-py27/5317715/ : SUCCESS in 12m 42s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/17/check/openstack-tox-py36/cae4111/ : SUCCESS in 21m 14s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/17/check/openstack-tox-py37/ef3c30e/ : SUCCESS in 12m 21s\n- openstack-tox-docs http://logs.openstack.org/81/639281/17/check/openstack-tox-docs/6679916/html/ : SUCCESS in 6m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/17/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8867169/ : SUCCESS in 45m 20s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/17/check/devstack-plugin-ceph-tempest/fd7c2d6/ : SUCCESS in 1h 46m 18s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/17/check/neutron-grenade-multinode/1ac4aed/ : SUCCESS in 1h 08m 13s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/17/check/neutron-tempest-linuxbridge/a543c53/ : SUCCESS in 1h 46m 57s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/17/check/nova-grenade-live-migration/63b2cb9/ : SUCCESS in 1h 05m 41s\n- nova-live-migration http://logs.openstack.org/81/639281/17/check/nova-live-migration/3b3cdb2/ : SUCCESS in 47m 58s\n- nova-lvm http://logs.openstack.org/81/639281/17/check/nova-lvm/4c8bea2/ : SUCCESS in 57m 35s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/17/check/nova-next/3482362/ : SUCCESS in 1h 29m 21s\n- nova-tox-functional http://logs.openstack.org/81/639281/17/check/nova-tox-functional/d5242a9/ : SUCCESS in 21m 40s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639281/17/check/nova-tox-functional-py36/0947275/ : SUCCESS in 17m 50s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/17/check/tempest-slow-py3/7fa1d94/ : SUCCESS in 2h 23m 12s","accounts_in_message":[],"_revision_number":17},{"id":"7d306383edd30438e35e798c92143678d7eca135","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 07:06:06.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/17/check-vote/ext-nova-zuul/6c90a6c : FAILURE in 1h 24m 52s","accounts_in_message":[],"_revision_number":17},{"id":"1c3ad357baefa84b19264866e951b4e53c10e493","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:30:51.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"ee4257fcd76a9f86e61c85817ad8a21a7d030bb7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:51:36.000000000","message":"Uploaded patch set 20: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"0d3ab23f4f6e84701151c419e05fc23a2c8b8e2b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-25 19:40:44.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/20/check/tempest-dsvm-full-xenial/020b1d4/ : SUCCESS in 1h 27m 11s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/20/check/tempest-dsvm-full-xenial-py3/3d31be2/ : SUCCESS in 1h 21m 09s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/20/check/grenade-dsvm-xenial/43e41f9/ : SUCCESS in 1h 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0f7bb5130c92a6af796ac2329eacea3cfaa27eb1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-25 20:36:24.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/20/check-vote/ext-nova-zuul/57defa5 : FAILURE in 14m 27s","accounts_in_message":[],"_revision_number":20},{"id":"d29dafc10a1ca51160d117ac91d360115bd429b3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-25 23:58:24.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/20/check/check-dsvm-tempest-vz7-exe-minimal/5c5c34e : FAILURE in 8m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"9f9dd54260f9bac4ac3b1b27d994decfd5713764","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 00:03:29.000000000","message":"Patch Set 20:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/20/check/nova-out-of-tree-pvm/a5acc9f : SUCCESS in 3h 02m 41s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/20/check/nova-in-tree-pvm/4631603 : FAILURE in 5h 01m 02s","accounts_in_message":[],"_revision_number":20},{"id":"80033f0dd4bfdc837fecafd4cc3a5a32d6b73a1b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 00:30:42.000000000","message":"Patch Set 20:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33156 : FAILURE in 3h 08m 45s","accounts_in_message":[],"_revision_number":20},{"id":"10de17188fe9ebc2349d74ecbd5ccce3dd599b13","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 00:53:15.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/20 : SUCCESS in 2h 17m 58s","accounts_in_message":[],"_revision_number":20},{"id":"fe3fa757c45976f20bdcc0462a5819111bc79b99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 06:53:22.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639281/20/check/grenade-py3/c0eda91/ : SUCCESS in 1h 02m 42s\n- tempest-full-py3 http://logs.openstack.org/81/639281/20/check/tempest-full-py3/e534191/ : SUCCESS in 1h 42m 36s\n- openstack-tox-cover http://logs.openstack.org/81/639281/20/check/openstack-tox-cover/d8241c9/ : FAILURE in 12m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/20/check/openstack-tox-lower-constraints/a2b7a4e/ : FAILURE in 13m 33s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/20/check/openstack-tox-pep8/2459820/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/20/check/openstack-tox-py27/e35383c/ : FAILURE in 12m 09s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/20/check/openstack-tox-py36/be737cf/ : FAILURE in 11m 25s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/20/check/openstack-tox-py37/fd6ead9/ : FAILURE in 10m 20s\n- openstack-tox-docs http://logs.openstack.org/81/639281/20/check/openstack-tox-docs/778c061/html/ : SUCCESS in 8m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/20/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1b9b1c0/ : FAILURE in 42m 09s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/20/check/devstack-plugin-ceph-tempest/d989a08/ : SUCCESS in 1h 23m 16s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/20/check/neutron-grenade-multinode/7456883/ : SUCCESS in 1h 26m 44s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/20/check/neutron-tempest-linuxbridge/9937475/ : SUCCESS in 1h 30m 39s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/20/check/nova-grenade-live-migration/d833029/ : SUCCESS in 1h 05m 30s\n- nova-live-migration http://logs.openstack.org/81/639281/20/check/nova-live-migration/85196fb/ : SUCCESS in 49m 12s\n- nova-lvm http://logs.openstack.org/81/639281/20/check/nova-lvm/99ee454/ : SUCCESS in 58m 21s (non-voting)\n- nova-multi-cell http://logs.openstack.org/81/639281/20/check/nova-multi-cell/fa7fc4b/ : SUCCESS in 1h 33m 26s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/20/check/nova-next/b1c6b70/ : SUCCESS in 1h 26m 24s\n- nova-tox-functional http://logs.openstack.org/81/639281/20/check/nova-tox-functional/1653f80/ : SUCCESS in 16m 47s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639281/20/check/nova-tox-functional-py36/ace41e8/ : SUCCESS in 16m 58s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/20/check/tempest-slow-py3/3bad474/ : SUCCESS in 2h 40m 04s","accounts_in_message":[],"_revision_number":20},{"id":"936fa0b4b159b447acbecfbb24b86255e4aac315","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-26 14:32:23.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"18ae0e3048e645a599dbbc571a1bf0118811e3cb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-26 16:20:25.000000000","message":"Patch Set 21:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/21/check/tempest-dsvm-full-xenial/9fe1e67/ : SUCCESS in 1h 27m 23s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/21/check/tempest-dsvm-full-xenial-py3/00e0026/ : SUCCESS in 1h 28m 45s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639281/21/check/grenade-dsvm-xenial/698af13/ : SUCCESS in 1h 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"0f8aa42939a0fd3fd0b02026c9860645324edbb4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-26 17:18:13.000000000","message":"Patch Set 21:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/21/check-vote/ext-nova-zuul/a631227 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":21},{"id":"d6755a91e64bc1f3fb1bc2a2af3a3df34205ca20","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 17:31:28.000000000","message":"Patch Set 21:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33197 : SUCCESS in 2h 49m 29s","accounts_in_message":[],"_revision_number":21},{"id":"8646a5bece7c0c05e80c97e7039343938b1c0601","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 19:20:31.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/21 : SUCCESS in 2h 33m 08s","accounts_in_message":[],"_revision_number":21},{"id":"a7c52ec400819f3dfee337f9e8468b3071106e25","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 19:59:44.000000000","message":"Patch Set 21:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639281/21/check/nova-out-of-tree-pvm/7b7fc8e : FAILURE in 5h 08m 53s\n- nova-in-tree-pvm http://184.172.12.213/81/639281/21/check/nova-in-tree-pvm/4ab9e86 : FAILURE in 4h 37m 33s","accounts_in_message":[],"_revision_number":21},{"id":"2b287ea1cfec971db62bde458e054174e792dca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:03:18.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/21/check/grenade-py3/be20a82/ : SUCCESS in 1h 05m 58s\n- tempest-full-py3 http://logs.openstack.org/81/639281/21/check/tempest-full-py3/d9c1d68/ : SUCCESS in 1h 46m 25s\n- openstack-tox-cover http://logs.openstack.org/81/639281/21/check/openstack-tox-cover/e21d661/cover/ : SUCCESS in 17m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/21/check/openstack-tox-lower-constraints/2b2e97f/ : SUCCESS in 12m 55s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/21/check/openstack-tox-pep8/6b8bb46/ : SUCCESS in 10m 12s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/21/check/openstack-tox-py27/95ac3b7/ : SUCCESS in 11m 07s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/21/check/openstack-tox-py36/2ea6712/ : SUCCESS in 10m 19s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/21/check/openstack-tox-py37/ac67c1d/ : SUCCESS in 12m 50s\n- openstack-tox-docs http://logs.openstack.org/81/639281/21/check/openstack-tox-docs/2640b61/html/ : SUCCESS in 5m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/21/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d17830f/ : SUCCESS in 53m 03s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/21/check/devstack-plugin-ceph-tempest/40b1249/ : SUCCESS in 1h 18m 04s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/21/check/neutron-grenade-multinode/a4a71eb/ : SUCCESS in 1h 14m 35s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/21/check/neutron-tempest-linuxbridge/461bfbe/ : SUCCESS in 1h 39m 27s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/21/check/nova-grenade-live-migration/e9ad45a/ : SUCCESS in 1h 07m 13s\n- nova-live-migration http://logs.openstack.org/81/639281/21/check/nova-live-migration/fa85c04/ : SUCCESS in 52m 23s\n- nova-lvm http://logs.openstack.org/81/639281/21/check/nova-lvm/2d0f1c5/ : SUCCESS in 1h 03m 30s (non-voting)\n- nova-multi-cell http://logs.openstack.org/81/639281/21/check/nova-multi-cell/e76baa2/ : SUCCESS in 1h 38m 14s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/21/check/nova-next/9751b6d/ : SUCCESS in 1h 23m 04s\n- nova-tox-functional http://logs.openstack.org/81/639281/21/check/nova-tox-functional/2c57eca/ : SUCCESS in 17m 22s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639281/21/check/nova-tox-functional-py36/16545f3/ : SUCCESS in 18m 22s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/21/check/tempest-slow-py3/7052c85/ : SUCCESS in 2h 38m 21s","accounts_in_message":[],"_revision_number":21},{"id":"4ebe70546a4899c5df872fcc179cd7459ff87a86","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 03:05:20.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/21/check/EMC_VxFlexOS_NOVA/bf60852/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":21},{"id":"cb607d192d173568151cfa1fee4d0c44ffa21062","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-27 04:43:18.000000000","message":"Patch Set 21:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-21 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":21},{"id":"8f0e16b026c89d4ec87114c7177244b4d576a62d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 13:05:45.000000000","message":"Patch Set 21:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/21/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/20d1f1b : SUCCESS in 2h 19m 40s","accounts_in_message":[],"_revision_number":21},{"id":"03408ea1f4bd146748fa95c3660c46955c159793","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-31 22:13:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"53f56a8706ba8eeec40264c46ebfa3ea25c04efa","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 22:17:14.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"9909a505442261c023108367483829ddc73f18d0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-31 23:44:06.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/22/check/tempest-dsvm-full-xenial/b1b87fc/ : SUCCESS in 1h 23m 07s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/22/check/tempest-dsvm-full-xenial-py3/5e7a098/ : SUCCESS in 1h 26m 59s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/22/check/grenade-dsvm-xenial/295f03b/ : SUCCESS in 1h 08m 59s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d385ea299bc241b24b898b1ee60aa049ba168efe","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 00:02:18.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/22/check/check-dsvm-tempest-vz7-exe-minimal/d210eef : FAILURE in 10m 46s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"687ff27c0c93d311cb0c3728579881a6bda7d721","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-01 00:20:59.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/639281/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"8a3c60f1d2a15949555b79f9f5b0d8777ea2c902","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-01 01:08:39.000000000","message":"Patch Set 22:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-34548 : FAILURE in 2h 48m 49s","accounts_in_message":[],"_revision_number":22},{"id":"3798033d720c01c1b47976e9f7a442a37472ee85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:25:33.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639281/22/check/grenade-py3/75337f0/ : SUCCESS in 1h 00m 23s\n- tempest-integrated-compute http://logs.openstack.org/81/639281/22/check/tempest-integrated-compute/c8d63d0/ : SUCCESS in 1h 21m 10s\n- openstack-tox-cover http://logs.openstack.org/81/639281/22/check/openstack-tox-cover/0a6c27e/cover/ : SUCCESS in 19m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639281/22/check/openstack-tox-lower-constraints/749a325/ : SUCCESS in 20m 26s\n- openstack-tox-pep8 http://logs.openstack.org/81/639281/22/check/openstack-tox-pep8/4c4ba8b/ : SUCCESS in 12m 33s\n- openstack-tox-py27 http://logs.openstack.org/81/639281/22/check/openstack-tox-py27/e66c7cf/ : SUCCESS in 22m 19s\n- openstack-tox-py36 http://logs.openstack.org/81/639281/22/check/openstack-tox-py36/c1b7c7f/ : SUCCESS in 14m 33s\n- openstack-tox-py37 http://logs.openstack.org/81/639281/22/check/openstack-tox-py37/20a73df/ : SUCCESS in 17m 11s\n- openstack-tox-docs http://logs.openstack.org/81/639281/22/check/openstack-tox-docs/1861162/html/ : SUCCESS in 5m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639281/22/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f9602b8/ : SUCCESS in 58m 14s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639281/22/check/devstack-plugin-ceph-tempest/7492aee/ : SUCCESS in 1h 11m 44s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639281/22/check/neutron-grenade-multinode/20485c5/ : SUCCESS in 1h 09m 43s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639281/22/check/neutron-tempest-linuxbridge/13113c2/ : SUCCESS in 1h 11m 21s\n- nova-grenade-live-migration http://logs.openstack.org/81/639281/22/check/nova-grenade-live-migration/31dd079/ : SUCCESS in 1h 02m 29s\n- nova-live-migration http://logs.openstack.org/81/639281/22/check/nova-live-migration/178dbf2/ : SUCCESS in 1h 03m 02s\n- nova-lvm http://logs.openstack.org/81/639281/22/check/nova-lvm/aff652e/ : SUCCESS in 47m 59s (non-voting)\n- nova-multi-cell http://logs.openstack.org/81/639281/22/check/nova-multi-cell/ca3c424/ : SUCCESS in 1h 28m 14s (non-voting)\n- nova-next http://logs.openstack.org/81/639281/22/check/nova-next/a74f01a/ : SUCCESS in 1h 31m 35s\n- nova-tox-functional http://logs.openstack.org/81/639281/22/check/nova-tox-functional/1801887/ : SUCCESS in 22m 10s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639281/22/check/nova-tox-functional-py36/bb146f5/ : SUCCESS in 21m 36s\n- tempest-slow-py3 http://logs.openstack.org/81/639281/22/check/tempest-slow-py3/8fc3387/ : SUCCESS in 2h 22m 56s\n- tempest-ipv6-only http://logs.openstack.org/81/639281/22/check/tempest-ipv6-only/f3ce535/ : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":22},{"id":"b9a82792c240fb9f6320c8fdac9b1987c68ab3b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-01 01:27:37.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639281/22/check-vote/ext-nova-zuul/229a963 : FAILURE in 1h 05m 59s","accounts_in_message":[],"_revision_number":22},{"id":"e3268afb59b3982755d857599386660d54bd0d00","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-01 02:59:55.000000000","message":"Patch Set 22:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-22 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":22},{"id":"2ada3df9664d446182e40983777bf4eec0bbe6a7","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-01 06:28:23.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/81/639281/22/check-nova/Nova-ML2-Sriov/24ba4cb : SUCCESS in 1h 25m 37s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"39a2c9078ee1d0df3fb0cfbbd3c01b94b27b6bd3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:27:37.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"b3aebbc0fa43b0e46d6fc3d5afd4951088e42ee3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:29:50.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"d4a63f7a249f98e55ce036a7e4328f3d0839b0a3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:30:32.000000000","message":"Patch Set 23:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639281/23/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/bd7a5a9 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":23},{"id":"443c08420ea9781c135aeb3c01e4e382a0edff5c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:39:16.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"ca9aac13dae55b603163c28c8b738746273c064b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-16 11:21:05.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/639281/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"6ed57907123f43abb9f505889f42ddc74e7dcbd2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-16 11:44:08.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35276 : FAILURE in 1h 09m 22s","accounts_in_message":[],"_revision_number":23},{"id":"018ad40dd5bca2c3cbf7436c42f4df463d3b04fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:56:29.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1fe45c72da194076b7f6be16ab73e529 : SUCCESS in 1h 02m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cde663c0ebb1447296588e5ff6d51783 : SUCCESS in 1h 24m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8523c25781e643a693c5ab34fbf4ef0f : SUCCESS in 19m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aa31e1b508341d2b056a9eee9c362ae : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba9561757c5488a844b31210715d073 : SUCCESS in 12m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20adba2c5db04952a3cbd416f700b7bb : SUCCESS in 17m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a881c41d63fa4a829fbaaa7e8f363397 : SUCCESS in 15m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a654952b5c54d54901ad1c204547410 : SUCCESS in 18m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82aeecd108344ccbfe8f8ec2ee9c871 : SUCCESS in 8m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/028fc2453f284ccf9eb5fd0dbd2a5e69 : SUCCESS in 54m 46s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbe68268c80c427c9e377f1bea0de324 : SUCCESS in 1h 19m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2129f5374c634797bd2b38145627f995 : SUCCESS in 1h 14m 21s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1d2be6239baa40babbfde22acc1be2be : SUCCESS in 1h 30m 29s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/6f14cc09e0ee4e5bb550b39efa3f03ac : SUCCESS in 1h 16m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1c2bed0092234d9a99803dcb566e3c40 : SUCCESS in 54m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/cf78a6b3005741619c7cdfc4794df64d : SUCCESS in 51m 51s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fd08445aa5664541af10a4eab2fb1dfd : SUCCESS in 1h 28m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/23e2fb544d6f4561b2b35e5abf20615d : SUCCESS in 1h 28m 45s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/6fedd14ca060408f9c237c5d5777b10e : SUCCESS in 21m 57s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9be8effd50854ab58a4260e5b15315c7 : SUCCESS in 17m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fec9711779f4c45aac3e9fed8930f1e : SUCCESS in 2h 21m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1238517a36f54107a04ed17352f5af01 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":23},{"id":"0cd6b7742fb28499df75dc8366c875552db1ab83","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 12:57:22.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639281/23/check/EMC_VxFlexOS_NOVA/87c9ebc/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"bcd0526a670620542c0a2bceebcc0ace41866f89","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-16 17:23:18.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/23 : SUCCESS in 2h 30m 12s","accounts_in_message":[],"_revision_number":23},{"id":"a296c263ef176f25a0d9bfbefb7f7beb373fc1e4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-16 23:03:21.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"6ec39fd639a6024332da10e920082f97b8a8d6a3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 01:57:50.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/23/check/check-dsvm-tempest-vz7-exe-minimal/25366ac : FAILURE in 11m 56s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"461b371c84f5d4d52285d8c5dfa32ee826d7bde3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 15:50:23.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"84650d6fa01b94354969832263a2865eaa9aac85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:50:53.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"37f59be21bd1ff9719026fcdb54b2d582cfa90b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 18:15:03.000000000","message":"Patch Set 23: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":23},{"id":"6dab52e6c1181292d07c6093ada80738275c691f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 18:18:07.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"188b0c1f15c0fa46d9c591573a83c1ef2163cf47","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 18:19:48.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f72e3941581213bc1a53cbfbb1e4c368b47412f6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-19 20:20:32.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35390 : FAILURE in 1h 28m 49s","accounts_in_message":[],"_revision_number":23},{"id":"42474446f7d7b25bcbd2e4d06ab128c2647463df","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-19 21:58:11.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639281/23/check/check-dsvm-tempest-vz7-exe-minimal/ed6630d : FAILURE in 1m 50s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"c0332b62734bdb50e760b881c16498f843f1a02a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-19 22:32:01.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639281/23 : SUCCESS in 2h 27m 38s","accounts_in_message":[],"_revision_number":23},{"id":"523b6d1c320aaf3a380bba2f0feeee34b53ce9e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:45:20.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f5cc38cd748f40b3b484358750b3f82f : SUCCESS in 1h 01m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5b8563c960ac4087bd7931a6cf2c60fc : SUCCESS in 1h 22m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28fe3dcca8ac42d587900680adbf0f45 : SUCCESS in 21m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b06d83132c324210b787d57622437dec : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2761ca25374e1f8568c914e2d114e2 : SUCCESS in 11m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fc9b23e336d476fa4db016a6177bec2 : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269ce3778ab4464e92ab4efc769f6c22 : SUCCESS in 11m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a54b8c88dd304b18a1f7079fa04e4971 : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d954b5f36040a2a9a701d18c6c1754 : SUCCESS in 7m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/91d2583448684244bc4b19deb813b290 : SUCCESS in 1h 02m 31s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5809ded4908f4c9e9b1fa364fb3521ed : SUCCESS in 1h 16m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/afee951827c64b3a9629c12079ca8640 : SUCCESS in 1h 23m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d9f8ab5b03fb498abfb3aa4ce42a4e83 : SUCCESS in 1h 14m 32s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/04252d9d561643bcb266b685400f17cd : SUCCESS in 1h 19m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cb8d09fbf34149fda1b8733765f47e3e : SUCCESS in 49m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/96c3a810496c400d97c242b0d7df46c1 : SUCCESS in 55m 09s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/794811f69c8c48458d9827bdae79e2e0 : SUCCESS in 1h 39m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/4d513258812b4ef383396156571186bb : SUCCESS in 1h 22m 29s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ca1f688ee28e4609a266d8083c691a27 : SUCCESS in 22m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a263fd4dd867474790339bf737cab052 : SUCCESS in 20m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e810ab2c1604737be35e71135f96ea5 : SUCCESS in 2h 39m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c4f1b4e8aa24e50a6d99e9e42f87b1b : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":23},{"id":"4249c6dbcf9297e9e0703da82411a18eb0100658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 22:45:52.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"2bd867583e4904709eec78d0e9fa9ebd07dfca1b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-19 22:57:55.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/23/check/tempest-dsvm-full-xenial/ccc45aa/ : SUCCESS in 1h 37m 23s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/23/check/tempest-dsvm-full-xenial-py3/c5eedc5/ : SUCCESS in 1h 36m 17s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639281/23/check/grenade-dsvm-xenial/c2e67d9/ : SUCCESS in 2h 02m 53s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"34230dd14b983f602864dfb4cdff70e1bc8d242d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-20 00:00:27.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639281-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"c691c4191724ee313782bda51d71e6bb7d3e06c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:56:59.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7933c092cfb44c9cba0f956318dcbe71 : SUCCESS in 1h 00m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0f3aaf3a8f024e849d76fbfab87b7ad7 : SUCCESS in 1h 22m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6748448211d431da1960ad826028c74 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57769653972b46fcb99d59522b95a005 : SUCCESS in 12m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d59f92ce930a415ab29fbc5bb1e20aba : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4260f1c5964282ad7432dbaed4625a : SUCCESS in 17m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f69b0a5535744f799b93b817c7e405e : SUCCESS in 17m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d97c7edef634bbbba446b00fad3f211 : SUCCESS in 9m 23s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/1ab57682c95c4c68bacf78e90d527fe1 : SUCCESS in 1h 13m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/257e21f3b3fd49798efb2baa276c37a2 : SUCCESS in 53m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ad6d308ee1d645308978e302bd79e692 : SUCCESS in 18m 43s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b834d1c3e96492a85c988624acebd6a : SUCCESS in 17m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3d8d4b6c2fcf4a9ca2bc14df3b207eda : SUCCESS in 1h 38m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/368c2b32431047c388b446dc129455d1 : SUCCESS in 1h 20m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9dd80a4a2f1c47e1823a556670bb7681 : SUCCESS in 2h 22m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf5b55e8611e4b9ea1605489e4c3d7a4 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":23},{"id":"ea962e6f6a90370aca36e4ee78bdd09f053ece27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:56:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"f2d93d09802bc3b9ae40e89088cce3fb34c47b52","revisions":{"cd6c272d25dbfc8032f2b69a0a7f7a4690f6cfb7":{"kind":"REWORK","_number":1,"created":"2019-02-26 09:43:48.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/1"}}},"commit":{"parents":[{"commit":"e750b24665b7afc7c30dde3524dec61bcca351de","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e750b24665b7afc7c30dde3524dec61bcca351de"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 09:43:31.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd6c272d25dbfc8032f2b69a0a7f7a4690f6cfb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd6c272d25dbfc8032f2b69a0a7f7a4690f6cfb7"}]},"branch":"refs/heads/master"},"597d2c8827c39cc409cc3b01003cb8e66f8219e7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-03-05 22:35:50.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/2"}}},"commit":{"parents":[{"commit":"bd37dfa5432ac543007fbcd8770809210579de76","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd37dfa5432ac543007fbcd8770809210579de76"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-05 22:18:10.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/597d2c8827c39cc409cc3b01003cb8e66f8219e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/597d2c8827c39cc409cc3b01003cb8e66f8219e7"}]},"branch":"refs/heads/master"},"359151cf49eff0eafcad0d7c52ca71c416a7a3d2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-06 19:09:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/639281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/3"}}},"commit":{"parents":[{"commit":"ba6c1b10c3a862c1bf5f5232f358742cb5956c17","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba6c1b10c3a862c1bf5f5232f358742cb5956c17"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-03-06 19:51:05.000000000","tz":-480},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/359151cf49eff0eafcad0d7c52ca71c416a7a3d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/359151cf49eff0eafcad0d7c52ca71c416a7a3d2"}]},"branch":"refs/heads/master"},"1ba46dedc373eb96addc6673376ee276738a71c1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-07 12:06:34.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/4"}}},"commit":{"parents":[{"commit":"31e2664fe2b9ffb6fc1672fffaa6374a0ba924b7","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/31e2664fe2b9ffb6fc1672fffaa6374a0ba924b7"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-07 12:05:58.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ba46dedc373eb96addc6673376ee276738a71c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ba46dedc373eb96addc6673376ee276738a71c1"}]},"branch":"refs/heads/master"},"23242ca8197d6fada0329fa0f5250435b96b57f0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-08 05:17:40.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/5"}}},"commit":{"parents":[{"commit":"1daf8db3c9fedc10351ccdafa52ef1cbbbee7a10","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1daf8db3c9fedc10351ccdafa52ef1cbbbee7a10"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-08 00:12:30.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23242ca8197d6fada0329fa0f5250435b96b57f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23242ca8197d6fada0329fa0f5250435b96b57f0"}]},"branch":"refs/heads/master"},"30d245379753706d3d57348b5f4eaf907f30ab24":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-18 05:44:49.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/6"}}},"commit":{"parents":[{"commit":"add02b042a0b4cdb78ad4eba828371276b4cf75a","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/add02b042a0b4cdb78ad4eba828371276b4cf75a"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-17 23:19:09.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30d245379753706d3d57348b5f4eaf907f30ab24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30d245379753706d3d57348b5f4eaf907f30ab24"}]},"branch":"refs/heads/master"},"24e64e7943f9d7fb9331d6daba3e17340eb8a1d4":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-21 09:37:18.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/7"}}},"commit":{"parents":[{"commit":"82faabca10b365a1daa5d70ca7d345fd919db346","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82faabca10b365a1daa5d70ca7d345fd919db346"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-21 09:37:01.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24e64e7943f9d7fb9331d6daba3e17340eb8a1d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24e64e7943f9d7fb9331d6daba3e17340eb8a1d4"}]},"branch":"refs/heads/master"},"982657c0286813954b54a2b01727e9845f24b880":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-03-27 04:38:51.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/8"}}},"commit":{"parents":[{"commit":"53d5d9d379beb9e5c673cbf41485a56e42d56105","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/53d5d9d379beb9e5c673cbf41485a56e42d56105"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-26 23:16:52.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/982657c0286813954b54a2b01727e9845f24b880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/982657c0286813954b54a2b01727e9845f24b880"}]},"branch":"refs/heads/master"},"319b329a4b1d29e838d83daf17ca8f2978483132":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-03-29 04:13:05.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/9"}}},"commit":{"parents":[{"commit":"2a5e8d7670c3079016ff2b285097ac39e5006c13","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a5e8d7670c3079016ff2b285097ac39e5006c13"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 04:10:28.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/319b329a4b1d29e838d83daf17ca8f2978483132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/319b329a4b1d29e838d83daf17ca8f2978483132"}]},"branch":"refs/heads/master"},"c346d19ec2605072619e4f6613648fce0d85dc1a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-29 06:09:26.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/10"}}},"commit":{"parents":[{"commit":"7611151bd975ff3e44979130ae4bd837b82f9db9","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7611151bd975ff3e44979130ae4bd837b82f9db9"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 06:04:21.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c346d19ec2605072619e4f6613648fce0d85dc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c346d19ec2605072619e4f6613648fce0d85dc1a"}]},"branch":"refs/heads/master"},"903547c5b072ceac4201ba218c9052e698eaa78b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-03-29 20:45:30.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/639281/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/11"}}},"commit":{"parents":[{"commit":"c78ef155c4f7621dbf026c58fc86cbb8533455d1","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c78ef155c4f7621dbf026c58fc86cbb8533455d1"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-03-29 20:41:16.000000000","tz":-300},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/903547c5b072ceac4201ba218c9052e698eaa78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/903547c5b072ceac4201ba218c9052e698eaa78b"}]},"branch":"refs/heads/master"},"2bfce7c82388ab5e50540b424b038b3abc533b2f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-01 21:18:06.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/12"}}},"commit":{"parents":[{"commit":"6da4b705d62ec7886a7f82964b869b0e0ee27eee","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6da4b705d62ec7886a7f82964b869b0e0ee27eee"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-01 21:15:55.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bfce7c82388ab5e50540b424b038b3abc533b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bfce7c82388ab5e50540b424b038b3abc533b2f"}]},"branch":"refs/heads/master"},"44839932b32c1fff6a39b8e455186a35de24e4de":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-04-02 10:13:10.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/13"}}},"commit":{"parents":[{"commit":"1d1acafe960e60a2c5fbb1e6769b8c1debc45b28","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d1acafe960e60a2c5fbb1e6769b8c1debc45b28"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-02 10:11:05.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44839932b32c1fff6a39b8e455186a35de24e4de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44839932b32c1fff6a39b8e455186a35de24e4de"}]},"branch":"refs/heads/master"},"882987b4c36d8e0140360252a72daec327262bb5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-04 08:44:44.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/14"}}},"commit":{"parents":[{"commit":"451397cc019eb264c2cbbbc2c6faa0a9444811bf","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/451397cc019eb264c2cbbbc2c6faa0a9444811bf"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 08:43:47.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/882987b4c36d8e0140360252a72daec327262bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/882987b4c36d8e0140360252a72daec327262bb5"}]},"branch":"refs/heads/master"},"a66bb1faa6e394af4c09c68ba65f5b749e968f84":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-04-04 20:44:48.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/15"}}},"commit":{"parents":[{"commit":"2b96b97ddaa512bcf89e550127fc5ad351e5da1f","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2b96b97ddaa512bcf89e550127fc5ad351e5da1f"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 20:44:02.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a66bb1faa6e394af4c09c68ba65f5b749e968f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a66bb1faa6e394af4c09c68ba65f5b749e968f84"}]},"branch":"refs/heads/master"},"3153d05c6fddb7c2423d608ee2a57bc30c690aa8":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-12 06:57:52.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/16"}}},"commit":{"parents":[{"commit":"841a4557610cdffa8712a956fd9dfd8682710f50","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/841a4557610cdffa8712a956fd9dfd8682710f50"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-11 22:22:08.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3153d05c6fddb7c2423d608ee2a57bc30c690aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3153d05c6fddb7c2423d608ee2a57bc30c690aa8"}]},"branch":"refs/heads/master"},"17017b1816a91af89700f368c2b67a3474146e70":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-04-24 01:44:56.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639281/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/17"}}},"commit":{"parents":[{"commit":"c222e9d38b130b831575b0cb30667224b9ae7220","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c222e9d38b130b831575b0cb30667224b9ae7220"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-23 21:50:56.000000000","tz":0},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17017b1816a91af89700f368c2b67a3474146e70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17017b1816a91af89700f368c2b67a3474146e70"}]},"branch":"refs/heads/master"},"ca4f99ddaabd438433ef5830d681a3472e154e95":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-06-25 16:30:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639281/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/18"}}},"commit":{"parents":[{"commit":"59a66a791b3b5f7a6d44a9ce288515f29e5a44e9","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59a66a791b3b5f7a6d44a9ce288515f29e5a44e9"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:24:49.000000000","tz":60},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca4f99ddaabd438433ef5830d681a3472e154e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca4f99ddaabd438433ef5830d681a3472e154e95"}]},"branch":"refs/heads/master"},"94c7e0e04781711063a11e4b8e6607b2383cb557":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-06-25 16:51:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639281/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/20"}}},"commit":{"parents":[{"commit":"ac7bca34ccabf620382801bed1fb082d37a3e7da","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac7bca34ccabf620382801bed1fb082d37a3e7da"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:51:20.000000000","tz":60},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94c7e0e04781711063a11e4b8e6607b2383cb557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94c7e0e04781711063a11e4b8e6607b2383cb557"}]},"branch":"refs/heads/master"},"b82b22f182dca6ab074fd4fd373d70c78cfa2f0e":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-06-26 14:32:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639281/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/21"}}},"commit":{"parents":[{"commit":"aec87ac53f7829e0ea57d3260f71382347ccf804","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aec87ac53f7829e0ea57d3260f71382347ccf804"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-26 14:30:20.000000000","tz":60},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b82b22f182dca6ab074fd4fd373d70c78cfa2f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b82b22f182dca6ab074fd4fd373d70c78cfa2f0e"}]},"branch":"refs/heads/master"},"4640d2b2022ed247830d2cd84fb368e8934a9362":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-07-31 22:13:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/639281/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/22"}}},"commit":{"parents":[{"commit":"5b7eced642141c2ef4a97b268722673b51954066","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b7eced642141c2ef4a97b268722673b51954066"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-31 22:13:27.000000000","tz":-300},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4640d2b2022ed247830d2cd84fb368e8934a9362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4640d2b2022ed247830d2cd84fb368e8934a9362"}]},"branch":"refs/heads/master"},"f2d93d09802bc3b9ae40e89088cce3fb34c47b52":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-08-16 10:27:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639281/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639281/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639281/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639281/23"}}},"commit":{"parents":[{"commit":"6e563d97030a960c1f042d2b908b4aa94ef214e1","subject":"Move dnsmasq restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e563d97030a960c1f042d2b908b4aa94ef214e1"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 00:49:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-16 10:24:25.000000000","tz":60},"subject":"Move router advertisement daemon restarts to privsep.","message":"Move router advertisement daemon restarts to privsep.\n\nA relatively simple case.\n\nChange-Id: I875483c869a253648fd90a9d3510f79cee509ce5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2d93d09802bc3b9ae40e89088cce3fb34c47b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2d93d09802bc3b9ae40e89088cce3fb34c47b52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
