)]}'
{"id":"openstack%2Fnova~639282","triplet_id":"openstack%2Fnova~master~I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822","project":"openstack/nova","branch":"master","topic":"my-own-personal-alternative-universe","hashtags":[],"change_id":"I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822","subject":"Move calls to ovs-vsctl to privsep.","status":"MERGED","created":"2019-02-26 09:43:48.000000000","updated":"2019-08-21 17:32:12.000000000","submitted":"2019-08-21 17:29:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639282-1566408597713-16970a7d","meta_rev_id":"91eccf8b1981031bad762cd6f3c419abe8a96fbe","_number":639282,"virtual_id_number":639282,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 13:01:14.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 13:22:41.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 14:01:48.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 10:12:11.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-08-21 11:22:15.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-08-20 13:33:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-08-21 17:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 15:52:42.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-08-21 12:25:11.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-08-21 06:58:52.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-08-20 19:14:00.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-08-20 19:14:00.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-06 19:09:40.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-03-27 09:47:08.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2019-06-26 21:03:39.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 02:20:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 10:40:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-08-16 11:40:02.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 12:48:39.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-08-20 13:01:14.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 13:22:41.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 13:33:06.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-08-20 15:52:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 19:14:00.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-21 06:58:52.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 10:12:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 11:22:15.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 12:25:11.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-21 14:01:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 17:29:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7589fb96ca507f2757f5a321f0bedd5abce46c76","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-02-26 09:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54daec7033178c4e1ca4e2aceff22f1ba1a19d42","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 10:16:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d847494f4c6c2db088c1026648539cec13dcb3a4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-26 12:35:48.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28270 : FAILURE in 1h 04m 47s","accounts_in_message":[],"_revision_number":1},{"id":"da87c3dc352da73881b02fe0fc74eb87cc404b26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 12:59:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/1/check/grenade-py3/845a85a/ : SUCCESS in 59m 29s\n- tempest-full-py3 http://logs.openstack.org/82/639282/1/check/tempest-full-py3/1cdf30b/ : SUCCESS in 1h 34m 56s\n- openstack-tox-cover http://logs.openstack.org/82/639282/1/check/openstack-tox-cover/4f16845/cover/ : SUCCESS in 18m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/1/check/openstack-tox-lower-constraints/89257a4/ : SUCCESS in 14m 59s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/1/check/openstack-tox-pep8/2fa722a/ : SUCCESS in 10m 53s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/1/check/openstack-tox-py27/3a88fca/ : SUCCESS in 12m 25s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/1/check/openstack-tox-py35/bd4fb46/ : SUCCESS in 13m 26s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/1/check/openstack-tox-py36/c7bf6d5/ : SUCCESS in 12m 20s\n- openstack-tox-docs http://logs.openstack.org/82/639282/1/check/openstack-tox-docs/0a7303d/html/ : SUCCESS in 7m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a88bec9/ : SUCCESS in 47m 13s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/1/check/devstack-plugin-ceph-tempest/10c9054/ : SUCCESS in 1h 56m 59s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/82/639282/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/6fecb32/ : SUCCESS in 1h 05m 15s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/1/check/neutron-grenade-multinode/f6a9085/ : SUCCESS in 1h 25m 26s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/1/check/neutron-tempest-linuxbridge/2f6aabe/ : SUCCESS in 1h 34m 32s\n- nova-live-migration http://logs.openstack.org/82/639282/1/check/nova-live-migration/34cdc5b/ : SUCCESS in 42m 55s\n- nova-lvm http://logs.openstack.org/82/639282/1/check/nova-lvm/c40e14d/ : SUCCESS in 1h 05m 49s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/1/check/nova-next/e8d484f/ : SUCCESS in 1h 56m 28s\n- nova-tox-functional http://logs.openstack.org/82/639282/1/check/nova-tox-functional/bdc8496/ : SUCCESS in 20m 42s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/1/check/nova-tox-functional-py35/767bd11/ : SUCCESS in 19m 15s\n- tempest-multinode-full http://logs.openstack.org/82/639282/1/check/tempest-multinode-full/3bb75c7/ : SUCCESS in 1h 43m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/1/check/tempest-slow-py3/7820f4c/ : SUCCESS in 2h 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6b6520abe2e3427cc4680cb0b98f7252d1b1fa01","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 13:02:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/1/check/EMC_VxFlexOS_NOVA/846a929/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"b76efebc92d926848d78d42123fe42660c550b1d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-26 13:03:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/1 : FAILURE in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b45e120008371403762893be62009d5b6982d435","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-26 15:31:31.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"e6162677ce311945092f45a48e6a91a4cbca32bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-26 19:01:00.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/1/check-vote/ext-nova-zuul/7bf0d20 : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b434a440986c61de8dd23e60860d99220d4742f1","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 10:42:06.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9307ee9 : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a63480f7a8cdc2a707a4bd40e6775716ba643051","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-27 14:59:13.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/1/check/tempest-dsvm-full-xenial/2a361d8/ : FAILURE in 23m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/1/check/tempest-dsvm-full-xenial-py3/fb52ef3/ : FAILURE in 18m 11s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/1/check/grenade-dsvm-xenial/9505586/ : FAILURE in 7m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d43f8937a150645964f1a410c5cf80e94690b060","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-05 22:35:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a2fb993e71c3df617e70de3e6410e26ded4e29ff","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-05 23:09:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"124ce420df0e58cad052d3da9350629005641227","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-06 00:13:53.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/2/check/nova-out-of-tree-pvm/23cf53f : FAILURE in 44m 55s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/2/check/nova-in-tree-pvm/75a51c6 : FAILURE in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"759bb33966d49811338612f333e0f20a245ad7f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 04:01:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/2/check/grenade-py3/6dc8afd/ : SUCCESS in 1h 00m 46s\n- tempest-full-py3 http://logs.openstack.org/82/639282/2/check/tempest-full-py3/3c4b007/ : FAILURE in 1h 12m 51s\n- openstack-tox-cover http://logs.openstack.org/82/639282/2/check/openstack-tox-cover/00dcac9/cover/ : SUCCESS in 17m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/2/check/openstack-tox-lower-constraints/5921c44/ : SUCCESS in 14m 57s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/2/check/openstack-tox-pep8/980d08d/ : SUCCESS in 10m 16s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/2/check/openstack-tox-py27/ff002c0/ : SUCCESS in 13m 15s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/2/check/openstack-tox-py35/9773658/ : SUCCESS in 12m 40s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/2/check/openstack-tox-py36/66e3985/ : SUCCESS in 11m 30s\n- openstack-tox-docs http://logs.openstack.org/82/639282/2/check/openstack-tox-docs/d8aff3d/html/ : SUCCESS in 6m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b87888d/ : SUCCESS in 49m 45s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/2/check/devstack-plugin-ceph-tempest/98c9319/ : SUCCESS in 1h 38m 23s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/82/639282/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/648c324/ : SUCCESS in 1h 08m 56s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/2/check/neutron-grenade-multinode/9287aca/ : SUCCESS in 1h 02m 51s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/2/check/neutron-tempest-linuxbridge/fa0071c/ : SUCCESS in 1h 22m 17s\n- nova-live-migration http://logs.openstack.org/82/639282/2/check/nova-live-migration/f74acc0/ : SUCCESS in 43m 39s\n- nova-lvm http://logs.openstack.org/82/639282/2/check/nova-lvm/d5f6382/ : FAILURE in 55m 33s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/2/check/nova-next/991bf97/ : FAILURE in 1h 48m 15s\n- nova-tox-functional http://logs.openstack.org/82/639282/2/check/nova-tox-functional/379bed8/ : SUCCESS in 22m 28s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/2/check/nova-tox-functional-py35/316f323/ : SUCCESS in 23m 04s\n- tempest-multinode-full http://logs.openstack.org/82/639282/2/check/tempest-multinode-full/3143e13/ : SUCCESS in 1h 54m 10s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/2/check/tempest-slow-py3/5f044ed/ : SUCCESS in 2h 27m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0bf65be646ba5820f7b4a2366964b098aa101b37","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 04:37:08.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28739 : SUCCESS in 1h 57m 06s","accounts_in_message":[],"_revision_number":2},{"id":"6748fb49f8958cef4912f306d0cde2d48aa41b53","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-06 06:11:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/2 : FAILURE in 2h 20m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c8e0b9a2cf856521ba5cdfbd9a81cc8958cbbc9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 13:39:07.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/2/check-vote/ext-nova-zuul/4f265b7 : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"57c98820f159a2ef827e235aa39d704df5e583a3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-03-06 19:09:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"47b0a7b1dd69ca5a1a6a13d1fbef34f1ce12d2d4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 22:24:48.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28826 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":3},{"id":"09a19ca36c3c1882eafc839d1acb5cf239f14fcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 00:29:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/3/check/grenade-py3/1dd751f/ : SUCCESS in 59m 50s\n- tempest-full-py3 http://logs.openstack.org/82/639282/3/check/tempest-full-py3/fc8e966/ : SUCCESS in 1h 33m 04s\n- openstack-tox-cover http://logs.openstack.org/82/639282/3/check/openstack-tox-cover/d82c925/cover/ : SUCCESS in 18m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/3/check/openstack-tox-lower-constraints/818fd37/ : SUCCESS in 13m 14s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/3/check/openstack-tox-pep8/663679d/ : SUCCESS in 11m 35s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/3/check/openstack-tox-py27/d3c5e3c/ : SUCCESS in 12m 53s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/3/check/openstack-tox-py35/331cc4b/ : SUCCESS in 14m 54s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/3/check/openstack-tox-py36/038c94e/ : SUCCESS in 13m 35s\n- openstack-tox-docs http://logs.openstack.org/82/639282/3/check/openstack-tox-docs/9d122c8/html/ : SUCCESS in 6m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b48df9e/ : SUCCESS in 46m 25s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/3/check/devstack-plugin-ceph-tempest/2d2d2cb/ : SUCCESS in 1h 39m 25s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/82/639282/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/1432951/ : SUCCESS in 1h 08m 44s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/3/check/neutron-grenade-multinode/818ee87/ : SUCCESS in 1h 08m 04s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/3/check/neutron-tempest-linuxbridge/b43543e/ : SUCCESS in 1h 39m 59s\n- nova-live-migration http://logs.openstack.org/82/639282/3/check/nova-live-migration/5bb449e/ : SUCCESS in 1h 02m 34s\n- nova-lvm http://logs.openstack.org/82/639282/3/check/nova-lvm/0909577/ : SUCCESS in 1h 07m 03s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/3/check/nova-next/739bd34/ : FAILURE in 1h 53m 41s\n- nova-tox-functional http://logs.openstack.org/82/639282/3/check/nova-tox-functional/03af9f3/ : SUCCESS in 26m 17s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/3/check/nova-tox-functional-py35/cf8151c/ : SUCCESS in 19m 40s\n- tempest-multinode-full http://logs.openstack.org/82/639282/3/check/tempest-multinode-full/ba41470/ : FAILURE in 1h 28m 15s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/3/check/tempest-slow-py3/8f635d2/ : SUCCESS in 2h 39m 59s","accounts_in_message":[],"_revision_number":3},{"id":"abf17000ee95e559ced3557baa9985889c5dc062","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 03:09:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/3 : SUCCESS in 3h 01m 20s","accounts_in_message":[],"_revision_number":3},{"id":"6cc309bc281d7e2597a60dfee988179e47cd2777","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-07 12:06:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"61d006b3314c7123df60a18ea1414fe3dd3fafa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 14:33:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/4/check/grenade-py3/45ecda5/ : SUCCESS in 1h 02m 49s\n- tempest-full-py3 http://logs.openstack.org/82/639282/4/check/tempest-full-py3/c5f48b3/ : SUCCESS in 1h 39m 06s\n- openstack-tox-cover http://logs.openstack.org/82/639282/4/check/openstack-tox-cover/9282148/ : FAILURE in 16m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/4/check/openstack-tox-lower-constraints/4a105a7/ : FAILURE in 15m 16s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/4/check/openstack-tox-pep8/5cc3ca9/ : FAILURE in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/4/check/openstack-tox-py27/609b969/ : FAILURE in 13m 18s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/4/check/openstack-tox-py35/a045d14/ : FAILURE in 12m 39s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/4/check/openstack-tox-py36/ee5fc70/ : FAILURE in 12m 17s\n- openstack-tox-docs http://logs.openstack.org/82/639282/4/check/openstack-tox-docs/1fab541/html/ : SUCCESS in 7m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/e8c76cb/ : SUCCESS in 57m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/4/check/devstack-plugin-ceph-tempest/3714472/ : SUCCESS in 1h 23m 23s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/82/639282/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/823015f/ : SUCCESS in 1h 12m 09s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/4/check/neutron-grenade-multinode/3c5c6a6/ : SUCCESS in 1h 14m 55s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/4/check/neutron-tempest-linuxbridge/2663795/ : SUCCESS in 1h 15m 12s\n- nova-live-migration http://logs.openstack.org/82/639282/4/check/nova-live-migration/28b13b7/ : SUCCESS in 52m 23s\n- nova-lvm http://logs.openstack.org/82/639282/4/check/nova-lvm/4f09633/ : SUCCESS in 1h 01m 59s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/4/check/nova-next/7a1f4ca/ : SUCCESS in 1h 50m 35s\n- nova-tox-functional http://logs.openstack.org/82/639282/4/check/nova-tox-functional/8d60441/ : SUCCESS in 19m 37s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/4/check/nova-tox-functional-py35/0ad92a9/ : SUCCESS in 19m 12s\n- tempest-multinode-full http://logs.openstack.org/82/639282/4/check/tempest-multinode-full/fcb199f/ : SUCCESS in 1h 45m 04s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/4/check/tempest-slow-py3/e35d648/ : FAILURE in 2h 18m 51s","accounts_in_message":[],"_revision_number":4},{"id":"eaf3086e55964735a96e2d38855056e98795b227","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-07 15:52:28.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28880 : FAILURE in 2h 50m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e2e47749cc87b968766e09bc9bbbbd26d1b39346","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 15:56:20.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/4 : FAILURE in 2h 18m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b6815d02fe88f2f8bff0f1dcb4c31375046cbcff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-07 22:54:56.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/4/check-vote/ext-nova-zuul/ce975e9 : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":4},{"id":"9d24c6a201deab5b2b275ff9635081fe28c18029","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 04:25:46.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ea45fc2 : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":4},{"id":"8c4dd10fc6363879c22805079d07e937c8619c07","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-08 05:17:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"20fe14beedc6ec3f215913405a289c80cd6cdb95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 07:37:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/5/check/grenade-py3/78cfe7c/ : SUCCESS in 1h 02m 21s\n- tempest-full-py3 http://logs.openstack.org/82/639282/5/check/tempest-full-py3/513022c/ : SUCCESS in 1h 26m 07s\n- openstack-tox-cover http://logs.openstack.org/82/639282/5/check/openstack-tox-cover/4745f3d/cover/ : SUCCESS in 17m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/5/check/openstack-tox-lower-constraints/df7fa0b/ : SUCCESS in 14m 26s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/5/check/openstack-tox-pep8/62b17ba/ : SUCCESS in 11m 12s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/5/check/openstack-tox-py27/1e65a56/ : SUCCESS in 13m 21s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/5/check/openstack-tox-py35/b77c626/ : SUCCESS in 13m 06s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/5/check/openstack-tox-py36/a4d88c6/ : SUCCESS in 13m 31s\n- openstack-tox-docs http://logs.openstack.org/82/639282/5/check/openstack-tox-docs/d3e384f/html/ : SUCCESS in 7m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8331c25/ : SUCCESS in 53m 18s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/5/check/devstack-plugin-ceph-tempest/c14430e/ : SUCCESS in 1h 47m 46s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/5/check/neutron-grenade-multinode/44d840d/ : SUCCESS in 1h 10m 55s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/5/check/neutron-tempest-linuxbridge/c76d663/ : SUCCESS in 1h 31m 47s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/5/check/nova-grenade-live-migration/d997bcf/ : SUCCESS in 1h 10m 31s\n- nova-live-migration http://logs.openstack.org/82/639282/5/check/nova-live-migration/a57822d/ : SUCCESS in 47m 03s\n- nova-lvm http://logs.openstack.org/82/639282/5/check/nova-lvm/f79c452/ : SUCCESS in 1h 00m 14s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/5/check/nova-next/45ef234/ : SUCCESS in 2h 07m 23s\n- nova-tox-functional http://logs.openstack.org/82/639282/5/check/nova-tox-functional/9cb29a0/ : SUCCESS in 18m 50s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/5/check/nova-tox-functional-py35/a8b510a/ : SUCCESS in 17m 43s\n- tempest-multinode-full http://logs.openstack.org/82/639282/5/check/tempest-multinode-full/014eae1/ : SUCCESS in 1h 55m 26s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/5/check/tempest-slow-py3/babdb61/ : SUCCESS in 2h 08m 39s","accounts_in_message":[],"_revision_number":5},{"id":"957b1f4cd086b6d3082a2f7ed0e1738605f06646","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 07:39:55.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/5/check/EMC_VxFlexOS_NOVA/8f136cc/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"25392a765159b0345109e1348d78eb2a94f1d953","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 08:55:57.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28940 : SUCCESS in 3h 07m 04s","accounts_in_message":[],"_revision_number":5},{"id":"c9af12a58e103ad9f4b21575b1657527db6ee46a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 15:07:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/5 : SUCCESS in 2h 25m 49s","accounts_in_message":[],"_revision_number":5},{"id":"041702b841ef283565ca814f73f9b6094c7ce986","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-03-08 17:28:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/82/639282/5/check-nova/Nova-ML2-Sriov/d4f76b8 : SUCCESS in 1h 10m 49s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/82/639282/5/check-nova/Nova-MACVTAP-ML2-Sriov/9fdc9f6 : SUCCESS in 1h 09m 24s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"7a1ff6627b8ec873d10fc92ddae9287b9f359f1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-08 23:02:53.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"6bd862b701db2d4cc0ec99b4d90a06f0ff85f9b6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-09 02:37:59.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/5/check-vote/ext-nova-zuul/1fef57c : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":5},{"id":"3bd37bfd29e0e30059e5cba90a40d839b622bb36","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 06:25:55.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5fa74dd : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":5},{"id":"245b7e1db2e462b9b3101e8e3cfd3f024281196c","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-18 05:44:49.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3890e17d513ddaf7d4df574ab170a3130ddeaf21","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 05:56:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89bf5277f3b2004fea089894c078528064e3baa4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-18 07:42:16.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29287 : SUCCESS in 1h 44m 48s","accounts_in_message":[],"_revision_number":6},{"id":"1642957927a008014f2b12c6a2436e925261641d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 08:11:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/6/check/grenade-py3/70c1aa4/ : SUCCESS in 59m 18s\n- tempest-full-py3 http://logs.openstack.org/82/639282/6/check/tempest-full-py3/9889d9a/ : SUCCESS in 1h 35m 16s\n- openstack-tox-cover http://logs.openstack.org/82/639282/6/check/openstack-tox-cover/2a83481/cover/ : SUCCESS in 15m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/6/check/openstack-tox-lower-constraints/411c0a4/ : SUCCESS in 14m 40s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/6/check/openstack-tox-pep8/56b0341/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/6/check/openstack-tox-py27/4840e3c/ : SUCCESS in 16m 42s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/6/check/openstack-tox-py35/6527ee4/ : SUCCESS in 14m 54s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/6/check/openstack-tox-py36/0eefd39/ : SUCCESS in 14m 15s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/6/check/openstack-tox-py37/60e09ae/ : SUCCESS in 13m 25s\n- openstack-tox-docs http://logs.openstack.org/82/639282/6/check/openstack-tox-docs/3452985/html/ : SUCCESS in 7m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/aa25fd6/ : SUCCESS in 52m 10s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/6/check/devstack-plugin-ceph-tempest/6a3620c/ : SUCCESS in 1h 43m 03s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/6/check/neutron-grenade-multinode/46cd5fc/ : SUCCESS in 1h 13m 58s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/6/check/neutron-tempest-linuxbridge/60a0654/ : SUCCESS in 1h 42m 53s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/6/check/nova-grenade-live-migration/0641264/ : SUCCESS in 1h 06m 05s\n- nova-live-migration http://logs.openstack.org/82/639282/6/check/nova-live-migration/00af706/ : SUCCESS in 56m 38s\n- nova-lvm http://logs.openstack.org/82/639282/6/check/nova-lvm/4980ab7/ : FAILURE in 58m 23s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/6/check/nova-next/6fafdab/ : FAILURE in 1h 22m 09s\n- nova-tox-functional http://logs.openstack.org/82/639282/6/check/nova-tox-functional/a10382a/ : SUCCESS in 20m 11s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/6/check/nova-tox-functional-py35/c669b54/ : SUCCESS in 19m 54s\n- tempest-multinode-full http://logs.openstack.org/82/639282/6/check/tempest-multinode-full/7ec9864/ : SUCCESS in 1h 38m 41s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/6/check/tempest-slow-py3/d732eb5/ : SUCCESS in 2h 15m 50s","accounts_in_message":[],"_revision_number":6},{"id":"59cc4e8c91d3262a57a65b7736d61d52b29c9c51","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 09:00:01.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/6/check-vote/ext-nova-zuul/e0384ee : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":6},{"id":"d53b8ee5eb0a19f5f4132caef434271bf57c9a59","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-18 10:17:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/6 : SUCCESS in 2h 34m 47s","accounts_in_message":[],"_revision_number":6},{"id":"8fb7f46b4fed5860ba3cd241af2824c4bd7af33c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 13:28:55.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/19e7758 : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":6},{"id":"c51eb42cfb1b2707978ee8c394ac969a6c1ccb2b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-18 19:30:15.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/6/check/tempest-dsvm-full-xenial/cbc80ce/ : SUCCESS in 1h 16m 30s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/6/check/tempest-dsvm-full-xenial-py3/e408018/ : SUCCESS in 1h 22m 04s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/6/check/grenade-dsvm-xenial/229a0d1/ : SUCCESS in 59m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0eed4d08507f5119e4f2c41f6b7cc63bb83754e7","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-21 09:37:18.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"6ad76c8fb436b4ce440c221e5142379b3fecf371","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 09:53:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7d74b6699d94960a10a7e7bfe79bb2fa37d4678e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-21 10:06:42.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29430 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":7},{"id":"bb7033d787a3546413588b55916ea4385988727e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:35:22.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/7/check/nova-out-of-tree-pvm/1e1b33d : FAILURE in 40m 26s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/7/check/nova-in-tree-pvm/eb424e2 : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":7},{"id":"86bbe698fc2c934fa4bd0e1941513567b03cdd75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-21 11:31:18.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/7/check/tempest-dsvm-full-xenial/d978684/ : SUCCESS in 1h 26m 43s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/7/check/tempest-dsvm-full-xenial-py3/7bc81ae/ : SUCCESS in 1h 11m 26s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/7/check/grenade-dsvm-xenial/823cfd6/ : SUCCESS in 58m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5c2c1f67d82541d0a5a77a9bc9dd09e3a675b453","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 12:11:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/7/check/grenade-py3/2843926/ : SUCCESS in 55m 49s\n- tempest-full-py3 http://logs.openstack.org/82/639282/7/check/tempest-full-py3/e420caa/ : SUCCESS in 1h 30m 46s\n- openstack-tox-cover http://logs.openstack.org/82/639282/7/check/openstack-tox-cover/4510408/cover/ : SUCCESS in 22m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/7/check/openstack-tox-lower-constraints/f991016/ : SUCCESS in 13m 35s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/7/check/openstack-tox-pep8/766e26b/ : SUCCESS in 10m 58s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/7/check/openstack-tox-py27/377e92c/ : SUCCESS in 14m 38s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/7/check/openstack-tox-py35/3d13831/ : SUCCESS in 13m 59s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/7/check/openstack-tox-py36/032ed7e/ : SUCCESS in 11m 57s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/7/check/openstack-tox-py37/0dc9cef/ : SUCCESS in 18m 31s\n- openstack-tox-docs http://logs.openstack.org/82/639282/7/check/openstack-tox-docs/0b7a285/html/ : SUCCESS in 7m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2924190/ : SUCCESS in 55m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/7/check/devstack-plugin-ceph-tempest/64b691e/ : SUCCESS in 1h 37m 08s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/7/check/neutron-grenade-multinode/3a5b581/ : SUCCESS in 1h 17m 09s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/7/check/neutron-tempest-linuxbridge/d066aa2/ : SUCCESS in 1h 29m 24s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/7/check/nova-grenade-live-migration/a4b9811/ : SUCCESS in 1h 09m 09s\n- nova-live-migration http://logs.openstack.org/82/639282/7/check/nova-live-migration/386c8fd/ : SUCCESS in 51m 29s\n- nova-lvm http://logs.openstack.org/82/639282/7/check/nova-lvm/009808e/ : SUCCESS in 1h 11m 22s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/7/check/nova-next/fd2961f/ : SUCCESS in 1h 25m 16s\n- nova-tox-functional http://logs.openstack.org/82/639282/7/check/nova-tox-functional/a01c8a4/ : SUCCESS in 22m 28s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/7/check/nova-tox-functional-py35/e1ff695/ : SUCCESS in 19m 33s\n- tempest-multinode-full http://logs.openstack.org/82/639282/7/check/tempest-multinode-full/54dabe2/ : SUCCESS in 1h 31m 23s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/82/639282/7/check/tempest-slow-py3/a1465ae/ : SUCCESS in 2h 14m 39s","accounts_in_message":[],"_revision_number":7},{"id":"353d199ba303e4d7a161315ce2b4ff86b6b843b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 12:12:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/7/check/EMC_VxFlexOS_NOVA/d99619e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"39f17c902255d3ddc920c195f81c66515e62e6f9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-21 12:22:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/7 : SUCCESS in 2h 28m 04s","accounts_in_message":[],"_revision_number":7},{"id":"64b79a41b9e7b591f62bd4ecfe672868a75e8ecd","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 16:41:25.000000000","message":"Patch Set 7:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/e2873c8 : FAILURE in 26m 49s","accounts_in_message":[],"_revision_number":7},{"id":"0e0d9d0313bff47a2affa456d2c64390e40c49d3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-21 20:04:50.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"c97154d549ef6dc34ecd09d9bec53dde46ef5da2","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-27 04:38:51.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"c6473b2658e01e4efa59d870fd2966aafc935dff","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 04:49:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"013077a4d48060eb8c20c789a1c9d230dd778c54","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-27 06:06:57.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/8/check/tempest-dsvm-full-xenial/d789b34/ : SUCCESS in 1h 11m 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/8/check/tempest-dsvm-full-xenial-py3/d01cdc6/ : SUCCESS in 1h 13m 42s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/8/check/grenade-dsvm-xenial/95cde6d/ : SUCCESS in 1h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c4cc289a1ffb59e096b82ff1dac80d597c44d479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 07:07:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/8/check/grenade-py3/9a2f4d1/ : SUCCESS in 57m 48s\n- tempest-full-py3 http://logs.openstack.org/82/639282/8/check/tempest-full-py3/9c16a15/ : SUCCESS in 1h 41m 04s\n- openstack-tox-cover http://logs.openstack.org/82/639282/8/check/openstack-tox-cover/c9925af/cover/ : SUCCESS in 18m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/8/check/openstack-tox-lower-constraints/e230056/ : SUCCESS in 16m 41s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/8/check/openstack-tox-pep8/ef43e33/ : SUCCESS in 9m 34s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/8/check/openstack-tox-py27/775a11f/ : SUCCESS in 15m 43s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/8/check/openstack-tox-py35/0f666b6/ : SUCCESS in 13m 45s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/8/check/openstack-tox-py36/fdbc443/ : SUCCESS in 16m 15s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/8/check/openstack-tox-py37/9454c11/ : SUCCESS in 15m 52s\n- openstack-tox-docs http://logs.openstack.org/82/639282/8/check/openstack-tox-docs/34ecb8b/html/ : SUCCESS in 8m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5bd6259/ : SUCCESS in 51m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/8/check/devstack-plugin-ceph-tempest/7f953f4/ : SUCCESS in 1h 46m 50s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/8/check/neutron-grenade-multinode/89b011f/ : SUCCESS in 1h 10m 36s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/8/check/neutron-tempest-linuxbridge/5181c65/ : SUCCESS in 1h 24m 38s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/8/check/nova-grenade-live-migration/2287bd3/ : SUCCESS in 1h 24m 40s\n- nova-live-migration http://logs.openstack.org/82/639282/8/check/nova-live-migration/ac648cd/ : SUCCESS in 47m 25s\n- nova-lvm http://logs.openstack.org/82/639282/8/check/nova-lvm/b35a181/ : SUCCESS in 1h 06m 26s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/8/check/nova-next/9f2d150/ : SUCCESS in 1h 36m 02s\n- nova-tox-functional http://logs.openstack.org/82/639282/8/check/nova-tox-functional/97016a9/ : SUCCESS in 22m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/8/check/nova-tox-functional-py35/ad2a464/ : SUCCESS in 25m 55s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/8/check/tempest-slow-py3/dcf2e37/ : SUCCESS in 2h 06m 21s","accounts_in_message":[],"_revision_number":8},{"id":"fc0c62dd79340235bbdb0939e772cba85032a764","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:07:51.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"4efab6e9ffbb666d6d28b432ff99f45ef02fa204","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:09:36.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/8/check/nova-out-of-tree-pvm/02cd050 : FAILURE in 2h 23m 55s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/8/check/nova-in-tree-pvm/49df3a1 : FAILURE in 1h 42m 11s","accounts_in_message":[],"_revision_number":8},{"id":"db3c03360dfa0cc5f0108586c5a6505c63267af0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:11:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/8/check/EMC_VxFlexOS_NOVA/e7283b4/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"65936bd3ef9747bd8ecd741f2d663b961d18a9fc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-27 07:28:57.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29582 : SUCCESS in 2h 37m 43s","accounts_in_message":[],"_revision_number":8},{"id":"03bc1e68915a6ad84f0bdfbd06bc1ed7b6d35ece","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-27 07:59:59.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/8 : FAILURE in 2h 38m 43s","accounts_in_message":[],"_revision_number":8},{"id":"60498ac15e72dcd10692a1f67939d649860ed886","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-03-27 09:47:08.000000000","message":"Patch Set 8:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":8},{"id":"f31491d6f2ba4e93fa35076a26490fbe3b0ddc5b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 10:55:03.000000000","message":"Patch Set 8:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/b0e2cc4 : SUCCESS in 1h 54m 32s","accounts_in_message":[],"_revision_number":8},{"id":"16af94c8c22bfa9027abc83d30667333745019b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-27 11:58:55.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"74fc3728c23f7fd314b7427c749fbc75822aa560","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 04:13:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"e0d7788d2f5f90838a6e176e751c5a85920bc655","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 04:32:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"701d8b949f2b37380504995117fdab35fd2e9299","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 05:59:48.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29660 : FAILURE in 1h 25m 28s","accounts_in_message":[],"_revision_number":9},{"id":"27a283151dd32b0b9e9a4be74c0363643e2e0412","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 06:06:07.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/9/check/tempest-dsvm-full-xenial/26abdf0/ : SUCCESS in 1h 18m 59s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/9/check/tempest-dsvm-full-xenial-py3/2d25acf/ : SUCCESS in 1h 15m 36s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/9/check/grenade-dsvm-xenial/16bb5fd/ : SUCCESS in 57m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"040e179536bd386354244db05b972e6221892c93","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 06:09:26.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"c4ad985d9b5e9bc48fe95a6b67af02d5e989b5bb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 06:34:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4960c6145badcdf2c47282d4e18251018b769e38","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 08:13:52.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/10/check/tempest-dsvm-full-xenial/19f59bc/ : SUCCESS in 1h 05m 12s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/10/check/tempest-dsvm-full-xenial-py3/0477622/ : SUCCESS in 1h 16m 47s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/10/check/grenade-dsvm-xenial/defca0d/ : SUCCESS in 1h 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d30477d5b4119f20499ef4ed88d4beaf70b145f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 08:57:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/10/check/grenade-py3/c0a901f/ : SUCCESS in 51m 57s\n- tempest-full-py3 http://logs.openstack.org/82/639282/10/check/tempest-full-py3/3525bc2/ : SUCCESS in 1h 24m 02s\n- openstack-tox-cover http://logs.openstack.org/82/639282/10/check/openstack-tox-cover/b88c1d0/cover/ : SUCCESS in 15m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/10/check/openstack-tox-lower-constraints/87caf26/ : SUCCESS in 14m 43s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/10/check/openstack-tox-pep8/e9d6481/ : FAILURE in 8m 55s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/10/check/openstack-tox-py27/912a69b/ : SUCCESS in 13m 23s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/10/check/openstack-tox-py35/937b3c3/ : SUCCESS in 13m 57s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/10/check/openstack-tox-py36/0fe59b3/ : SUCCESS in 14m 29s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/10/check/openstack-tox-py37/beb4fc1/ : SUCCESS in 12m 52s\n- openstack-tox-docs http://logs.openstack.org/82/639282/10/check/openstack-tox-docs/8e5184e/html/ : SUCCESS in 8m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b9743c4/ : SUCCESS in 46m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/10/check/devstack-plugin-ceph-tempest/a9a1ec7/ : SUCCESS in 1h 21m 59s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/10/check/neutron-grenade-multinode/81ceaa2/ : SUCCESS in 1h 04m 03s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/10/check/neutron-tempest-linuxbridge/4013617/ : SUCCESS in 1h 17m 05s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/10/check/nova-grenade-live-migration/5415ff7/ : SUCCESS in 1h 03m 09s\n- nova-live-migration http://logs.openstack.org/82/639282/10/check/nova-live-migration/46c192e/ : SUCCESS in 45m 55s\n- nova-lvm http://logs.openstack.org/82/639282/10/check/nova-lvm/5b46820/ : SUCCESS in 57m 17s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/10/check/nova-next/019e6d7/ : SUCCESS in 1h 20m 46s\n- nova-tox-functional http://logs.openstack.org/82/639282/10/check/nova-tox-functional/0459422/ : SUCCESS in 19m 31s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/10/check/nova-tox-functional-py35/5f18353/ : SUCCESS in 20m 41s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/10/check/tempest-slow-py3/c9aea51/ : SUCCESS in 2h 17m 46s","accounts_in_message":[],"_revision_number":10},{"id":"c8fbb8a8e091898c8c05779fdd221348f91a1dcf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 09:38:02.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29679 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":10},{"id":"1361dbb0a17fdbc3e5f024c083ed182252545fb3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 10:58:14.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/10 : FAILURE in 2h 13m 12s","accounts_in_message":[],"_revision_number":10},{"id":"96ec2111976aa53b59b30fc1edc7ec56ceaf5089","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-29 10:59:33.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/10/check/nova-out-of-tree-pvm/bd55b37 : FAILURE in 2h 51m 12s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/10/check/nova-in-tree-pvm/0487f81 : FAILURE in 3h 00m 55s","accounts_in_message":[],"_revision_number":10},{"id":"77b85a75add188db9ca392e3dfa75e26740f8d54","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 13:04:20.000000000","message":"Patch Set 10:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/55bde3a : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":10},{"id":"9f22d703c65328a6beaee3607ee6ec7756bf6c3f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 20:45:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"2352a2184ea79ba70d9b3c12b6affa851e0e0077","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 21:06:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c63baf10c750254547d6ce939ff7d15c54fb0ddd","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 21:57:12.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29721 : FAILURE in 24m 10s","accounts_in_message":[],"_revision_number":11},{"id":"45c0f1be895dc7b7ce4ae0e8a3fe3f85db315b6e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 22:49:15.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/11/check/tempest-dsvm-full-xenial/cb4e492/ : SUCCESS in 1h 11m 53s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/11/check/tempest-dsvm-full-xenial-py3/111f271/ : SUCCESS in 1h 19m 32s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/11/check/grenade-dsvm-xenial/3250fb9/ : SUCCESS in 1h 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"478dc1b0eeb4b03d01fada888d24ba4385a51404","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 23:19:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/11/check/grenade-py3/822307e/ : FAILURE in 37m 35s\n- tempest-full-py3 http://logs.openstack.org/82/639282/11/check/tempest-full-py3/4dc6186/ : SUCCESS in 1h 30m 32s\n- openstack-tox-cover http://logs.openstack.org/82/639282/11/check/openstack-tox-cover/ea21beb/cover/ : SUCCESS in 16m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/11/check/openstack-tox-lower-constraints/cc94caa/ : SUCCESS in 14m 42s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/11/check/openstack-tox-pep8/cf14da9/ : SUCCESS in 10m 07s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/11/check/openstack-tox-py27/f93d7f9/ : SUCCESS in 12m 31s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/11/check/openstack-tox-py35/4d1b809/ : SUCCESS in 12m 17s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/11/check/openstack-tox-py36/e9a20ac/ : SUCCESS in 12m 08s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/11/check/openstack-tox-py37/eb399d1/ : SUCCESS in 12m 33s\n- openstack-tox-docs http://logs.openstack.org/82/639282/11/check/openstack-tox-docs/e5903ee/html/ : SUCCESS in 6m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/484e7d3/ : SUCCESS in 50m 31s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/11/check/devstack-plugin-ceph-tempest/0d2f511/ : SUCCESS in 1h 30m 44s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/11/check/neutron-grenade-multinode/6c115bf/ : FAILURE in 54m 47s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/11/check/neutron-tempest-linuxbridge/4f7118a/ : SUCCESS in 1h 32m 03s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/11/check/nova-grenade-live-migration/788ccbb/ : SUCCESS in 1h 07m 46s\n- nova-live-migration http://logs.openstack.org/82/639282/11/check/nova-live-migration/a4e45b2/ : SUCCESS in 55m 44s\n- nova-lvm http://logs.openstack.org/82/639282/11/check/nova-lvm/1cad8e4/ : SUCCESS in 1h 01m 53s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/11/check/nova-next/6a19730/ : SUCCESS in 1h 20m 30s\n- nova-tox-functional http://logs.openstack.org/82/639282/11/check/nova-tox-functional/fc5cefc/ : SUCCESS in 19m 59s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/11/check/nova-tox-functional-py35/246ad45/ : SUCCESS in 19m 10s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/11/check/tempest-slow-py3/c2ca9e2/ : SUCCESS in 2h 19m 47s","accounts_in_message":[],"_revision_number":11},{"id":"a18be3717c2e682df986b06cf1bb5224945a2a41","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-30 00:17:28.000000000","message":"Patch Set 11:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/11/check/nova-out-of-tree-pvm/cc52e40 : SUCCESS in 2h 27m 32s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/11/check/nova-in-tree-pvm/0914130 : FAILURE in 2h 18m 55s","accounts_in_message":[],"_revision_number":11},{"id":"229938c1bcd3ea296f8cd8ad8e39f849906c402b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-30 01:23:11.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/11 : FAILURE in 2h 15m 18s","accounts_in_message":[],"_revision_number":11},{"id":"ba70b98c0f461cc2dd0cd93a69e445152f0ea71a","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 15:02:25.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/cd6beb3 : SUCCESS in 1h 34m 52s","accounts_in_message":[],"_revision_number":11},{"id":"83bfd0212a85d93152bad73f5976b865e845eb0a","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-01 21:18:06.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"d2cfc89101cdcadf320f69796fb52affcacace34","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 21:44:59.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"213efe6bc5c5b3d85f9974babf55e5ee25b3019d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-01 23:30:40.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/12/check/tempest-dsvm-full-xenial/fbfe5df/ : SUCCESS in 1h 21m 08s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/12/check/tempest-dsvm-full-xenial-py3/d104333/ : SUCCESS in 1h 06m 21s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/12/check/grenade-dsvm-xenial/b2c2e09/ : SUCCESS in 58m 46s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"81d40911320ed1bdde6c037b794500a7b01dbfac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 00:21:02.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/12/check/grenade-py3/f36647a/ : SUCCESS in 1h 06m 47s\n- tempest-full-py3 http://logs.openstack.org/82/639282/12/check/tempest-full-py3/d62e528/ : SUCCESS in 1h 21m 16s\n- openstack-tox-cover http://logs.openstack.org/82/639282/12/check/openstack-tox-cover/17eb1df/ : FAILURE in 12m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/12/check/openstack-tox-lower-constraints/8a30b4c/ : FAILURE in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/12/check/openstack-tox-pep8/ad13cd1/ : SUCCESS in 10m 42s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/12/check/openstack-tox-py27/b950d08/ : SUCCESS in 14m 57s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/12/check/openstack-tox-py35/74c6aee/ : FAILURE in 14m 45s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/12/check/openstack-tox-py36/ad11554/ : FAILURE in 13m 08s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/12/check/openstack-tox-py37/afec172/ : FAILURE in 13m 17s\n- openstack-tox-docs http://logs.openstack.org/82/639282/12/check/openstack-tox-docs/aa7c77e/html/ : SUCCESS in 6m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/36fad59/ : SUCCESS in 56m 37s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/12/check/devstack-plugin-ceph-tempest/d797bc9/ : SUCCESS in 1h 29m 03s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/12/check/neutron-grenade-multinode/701c197/ : SUCCESS in 1h 14m 45s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/12/check/neutron-tempest-linuxbridge/3f050d5/ : SUCCESS in 1h 34m 59s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/12/check/nova-grenade-live-migration/7dca022/ : SUCCESS in 1h 08m 50s\n- nova-live-migration http://logs.openstack.org/82/639282/12/check/nova-live-migration/22b7978/ : FAILURE in 41m 39s\n- nova-lvm http://logs.openstack.org/82/639282/12/check/nova-lvm/d61c761/ : SUCCESS in 1h 06m 15s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/12/check/nova-next/0bb538e/ : SUCCESS in 1h 23m 48s\n- nova-tox-functional http://logs.openstack.org/82/639282/12/check/nova-tox-functional/2edd921/ : SUCCESS in 19m 47s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/12/check/nova-tox-functional-py35/d889227/ : SUCCESS in 24m 34s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/12/check/tempest-slow-py3/deb1cd4/ : SUCCESS in 2h 21m 41s","accounts_in_message":[],"_revision_number":12},{"id":"f59f44ce3df83015551af01c3fc7d69e719da07e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 00:46:42.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29796 : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":12},{"id":"58e2d71975d36161c8f6b8b2509c715b40e56ef8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 02:32:22.000000000","message":"Patch Set 12:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c62fe4f : FAILURE in 20m 50s","accounts_in_message":[],"_revision_number":12},{"id":"c2acc5851518c1f6318795fdbdb2fc7eec2a20ad","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 02:35:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/12 : SUCCESS in 2h 37m 47s","accounts_in_message":[],"_revision_number":12},{"id":"c42e8072d4d5a9a0a931f961b18a549bac612ed8","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-02 10:13:10.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"14ace5d6f389657c5b7e2685f417c15435744c72","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 10:45:18.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7458920cce8daaccdae3643d7dc957f51decd837","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 10:51:15.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/13 : FAILURE in 45s","accounts_in_message":[],"_revision_number":13},{"id":"29c5b86c4b9624b38382c02a40803a9dcd11458e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 13:29:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/13/check/grenade-py3/659fc68/ : SUCCESS in 58m 26s\n- tempest-full-py3 http://logs.openstack.org/82/639282/13/check/tempest-full-py3/edd2ee9/ : SUCCESS in 1h 39m 30s\n- openstack-tox-cover http://logs.openstack.org/82/639282/13/check/openstack-tox-cover/b233ba9/ : FAILURE in 12m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/13/check/openstack-tox-lower-constraints/c1b3def/ : FAILURE in 18m 01s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/13/check/openstack-tox-pep8/1d1beb7/ : SUCCESS in 15m 33s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/13/check/openstack-tox-py27/640993c/ : SUCCESS in 15m 03s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/13/check/openstack-tox-py35/31c6b8f/ : FAILURE in 13m 39s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/13/check/openstack-tox-py36/262d2db/ : FAILURE in 11m 36s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/13/check/openstack-tox-py37/7c8630f/ : FAILURE in 12m 58s\n- openstack-tox-docs http://logs.openstack.org/82/639282/13/check/openstack-tox-docs/623a1cd/html/ : SUCCESS in 7m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/13/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/7e39c17/ : SUCCESS in 51m 56s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/13/check/devstack-plugin-ceph-tempest/f2bd12c/ : SUCCESS in 1h 33m 02s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/13/check/neutron-grenade-multinode/40bbcdc/ : SUCCESS in 1h 08m 54s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/13/check/neutron-tempest-linuxbridge/9f23dc5/ : SUCCESS in 1h 48m 26s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/13/check/nova-grenade-live-migration/000593d/ : SUCCESS in 1h 25m 13s\n- nova-live-migration http://logs.openstack.org/82/639282/13/check/nova-live-migration/073f21b/ : SUCCESS in 54m 47s\n- nova-lvm http://logs.openstack.org/82/639282/13/check/nova-lvm/a7f39df/ : SUCCESS in 1h 15m 09s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/13/check/nova-next/b70d706/ : SUCCESS in 1h 26m 42s\n- nova-tox-functional http://logs.openstack.org/82/639282/13/check/nova-tox-functional/85ab2d2/ : SUCCESS in 24m 15s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/13/check/nova-tox-functional-py35/ab74f26/ : SUCCESS in 23m 51s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/13/check/tempest-slow-py3/7f1e53b/ : SUCCESS in 2h 17m 07s","accounts_in_message":[],"_revision_number":13},{"id":"1acbe281ccfb2bd3c53bb8ba2d26e324ef329716","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 15:56:18.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29885 : SUCCESS in 1h 55m 37s","accounts_in_message":[],"_revision_number":13},{"id":"2f6c7bd5b97e715b14819dc4b67730334710fa24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-02 20:25:27.000000000","message":"Patch Set 13:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/13/check/tempest-dsvm-full-xenial/a5dfab6/ : SUCCESS in 1h 24m 09s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/13/check/tempest-dsvm-full-xenial-py3/afd187d/ : SUCCESS in 1h 14m 48s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/13/check/grenade-dsvm-xenial/dc8e7cc/ : SUCCESS in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4345ca1242c55157c5bc0daef082308de6d2d7a9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-03 09:46:07.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7f4f9c4 : SUCCESS in 2h 17m 54s","accounts_in_message":[],"_revision_number":13},{"id":"c88c349323d0b4b6ecc990292f060dd0c958fe51","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 08:44:44.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"6bad5eac93e68b43daf16bc75efdcbffcf1b6d43","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-04 09:13:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"26d3d4a1543fba3476079f58695b243ac8f0c67e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 12:25:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/14/check/grenade-py3/23b836a/ : SUCCESS in 1h 00m 27s\n- tempest-full-py3 http://logs.openstack.org/82/639282/14/check/tempest-full-py3/3c4bc80/ : SUCCESS in 1h 23m 39s\n- openstack-tox-cover http://logs.openstack.org/82/639282/14/check/openstack-tox-cover/e7dfb42/ : FAILURE in 14m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/14/check/openstack-tox-lower-constraints/967924e/ : FAILURE in 14m 49s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/14/check/openstack-tox-pep8/3bc6ae6/ : SUCCESS in 9m 37s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/14/check/openstack-tox-py27/16af65c/ : SUCCESS in 13m 37s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/14/check/openstack-tox-py35/8817b69/ : FAILURE in 14m 59s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/14/check/openstack-tox-py36/dff3fbb/ : FAILURE in 12m 13s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/14/check/openstack-tox-py37/7207612/ : FAILURE in 12m 17s\n- openstack-tox-docs http://logs.openstack.org/82/639282/14/check/openstack-tox-docs/9a8fc23/html/ : SUCCESS in 6m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/14/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c97fc63/ : TIMED_OUT in 1h 33m 49s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/14/check/devstack-plugin-ceph-tempest/2b20064/ : SUCCESS in 1h 33m 26s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/14/check/neutron-grenade-multinode/7a6ea01/ : SUCCESS in 1h 07m 37s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/14/check/neutron-tempest-linuxbridge/407d03e/ : SUCCESS in 1h 42m 32s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/14/check/nova-grenade-live-migration/e80b9be/ : SUCCESS in 1h 13m 06s\n- nova-live-migration http://logs.openstack.org/82/639282/14/check/nova-live-migration/8ff956e/ : SUCCESS in 46m 06s\n- nova-lvm http://logs.openstack.org/82/639282/14/check/nova-lvm/fef5c47/ : SUCCESS in 1h 24m 12s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/14/check/nova-next/e833b49/ : SUCCESS in 1h 50m 27s\n- nova-tox-functional http://logs.openstack.org/82/639282/14/check/nova-tox-functional/fa0e468/ : SUCCESS in 31m 33s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/14/check/nova-tox-functional-py35/d8a4cac/ : SUCCESS in 20m 25s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/14/check/tempest-slow-py3/f74a560/ : SUCCESS in 2h 18m 22s","accounts_in_message":[],"_revision_number":14},{"id":"e6e1f3a4a453023730c7bbd8a64673a882c4b56c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-04 16:12:13.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30033 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":14},{"id":"e7140094725e2020bbd48ea6ec832c02cf070621","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 18:16:39.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/14/check/tempest-dsvm-full-xenial/6a59f99/ : SUCCESS in 1h 25m 18s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/14/check/tempest-dsvm-full-xenial-py3/3bf8055/ : SUCCESS in 1h 23m 31s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/14/check/grenade-dsvm-xenial/70f37cc/ : SUCCESS in 1h 00m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f53db6574809e7382138bdde4601956f64107ac9","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 20:44:48.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"dfa52d706f6ff15e8edc8825882ee628d0f1ae8d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 22:58:07.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/15/check/tempest-dsvm-full-xenial/438cdcb/ : SUCCESS in 1h 17m 41s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/15/check/tempest-dsvm-full-xenial-py3/23ac87c/ : SUCCESS in 1h 16m 19s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/15/check/grenade-dsvm-xenial/bc3b337/ : SUCCESS in 1h 02m 15s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"15f48f68bb54c9959c5ed3aab62eca2756e66c31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 00:02:54.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/15/check/grenade-py3/5cdb523/ : SUCCESS in 54m 55s\n- tempest-full-py3 http://logs.openstack.org/82/639282/15/check/tempest-full-py3/0c3c816/ : FAILURE in 1h 10m 18s\n- openstack-tox-cover http://logs.openstack.org/82/639282/15/check/openstack-tox-cover/79e0198/cover/ : SUCCESS in 16m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/15/check/openstack-tox-lower-constraints/33cb6b6/ : SUCCESS in 17m 03s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/15/check/openstack-tox-pep8/08632f9/ : SUCCESS in 11m 38s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/15/check/openstack-tox-py27/04dbdec/ : SUCCESS in 14m 28s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/15/check/openstack-tox-py35/b1ef74e/ : SUCCESS in 16m 21s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/15/check/openstack-tox-py36/c6a78fb/ : SUCCESS in 12m 48s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/15/check/openstack-tox-py37/fb40abe/ : SUCCESS in 14m 04s\n- openstack-tox-docs http://logs.openstack.org/82/639282/15/check/openstack-tox-docs/4fc2666/html/ : SUCCESS in 7m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/15/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/674fa1e/ : SUCCESS in 49m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/15/check/devstack-plugin-ceph-tempest/f1aaa2c/ : SUCCESS in 1h 42m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/15/check/neutron-grenade-multinode/969a916/ : SUCCESS in 1h 06m 16s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/15/check/neutron-tempest-linuxbridge/1c48abd/ : SUCCESS in 1h 25m 35s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/15/check/nova-grenade-live-migration/a648612/ : FAILURE in 45m 14s\n- nova-live-migration http://logs.openstack.org/82/639282/15/check/nova-live-migration/f2e9246/ : SUCCESS in 52m 19s\n- nova-lvm http://logs.openstack.org/82/639282/15/check/nova-lvm/bd55033/ : SUCCESS in 1h 03m 09s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/15/check/nova-next/140c9bd/ : SUCCESS in 1h 37m 03s\n- nova-tox-functional http://logs.openstack.org/82/639282/15/check/nova-tox-functional/f140217/ : SUCCESS in 22m 48s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/15/check/nova-tox-functional-py35/199e173/ : SUCCESS in 24m 56s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/15/check/tempest-slow-py3/ee48948/ : SUCCESS in 2h 23m 20s","accounts_in_message":[],"_revision_number":15},{"id":"bcd570ccae51810a4120aaff5f2fad0c676c725c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-05 00:33:52.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30074 : SUCCESS in 1h 52m 57s","accounts_in_message":[],"_revision_number":15},{"id":"75a4f487bf7064a0e776b9f55578bad5890897ff","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-05 02:21:00.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/15 : FAILURE in 2h 16m 06s","accounts_in_message":[],"_revision_number":15},{"id":"49eefed962af5125b7976e133628364056304d63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-05 15:33:36.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/15/check-vote/ext-nova-zuul/ecf4ad5 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":15},{"id":"8646f35877e923c3e4fb085b251b1890b18e648b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-07 05:33:12.000000000","message":"Patch Set 15:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/840a92e : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":15},{"id":"3a11bf0cf9a9d02de152016e94afdc20802c5f71","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-08 01:08:25.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/15/check/check-dsvm-tempest-vz7-exe-minimal/e89529f : FAILURE in 1h 07m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"2e8d2145b892c59d9895602d4d1fbd2983504a46","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-12 06:57:52.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"4f2ddfb55e1f139b6de228e9d6772a64baee0388","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 07:13:29.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"5b576652e1339cd991a1acb3f9f2f2072ac173a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 09:52:45.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/16/check/grenade-py3/750b47f/ : SUCCESS in 59m 34s\n- tempest-full-py3 http://logs.openstack.org/82/639282/16/check/tempest-full-py3/8be0162/ : SUCCESS in 1h 32m 00s\n- openstack-tox-cover http://logs.openstack.org/82/639282/16/check/openstack-tox-cover/d67259a/cover/ : SUCCESS in 16m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/16/check/openstack-tox-lower-constraints/742862f/ : SUCCESS in 15m 06s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/16/check/openstack-tox-pep8/9912857/ : SUCCESS in 10m 47s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/16/check/openstack-tox-py27/f600bdf/ : SUCCESS in 15m 47s\n- openstack-tox-py35 http://logs.openstack.org/82/639282/16/check/openstack-tox-py35/78671fa/ : SUCCESS in 14m 51s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/16/check/openstack-tox-py36/365a4c3/ : SUCCESS in 16m 43s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/16/check/openstack-tox-py37/5b219fc/ : SUCCESS in 14m 18s\n- openstack-tox-docs http://logs.openstack.org/82/639282/16/check/openstack-tox-docs/a11103e/html/ : SUCCESS in 9m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/16/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/64708c9/ : SUCCESS in 52m 26s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/16/check/devstack-plugin-ceph-tempest/a026992/ : SUCCESS in 1h 24m 51s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/16/check/neutron-grenade-multinode/886066b/ : SUCCESS in 1h 03m 50s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/16/check/neutron-tempest-linuxbridge/b6c1e38/ : SUCCESS in 1h 26m 27s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/16/check/nova-grenade-live-migration/7cd2cb3/ : SUCCESS in 1h 08m 33s\n- nova-live-migration http://logs.openstack.org/82/639282/16/check/nova-live-migration/05ea4bf/ : SUCCESS in 54m 18s\n- nova-lvm http://logs.openstack.org/82/639282/16/check/nova-lvm/77adfa8/ : SUCCESS in 1h 13m 13s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/16/check/nova-next/07753bd/ : SUCCESS in 1h 35m 47s\n- nova-tox-functional http://logs.openstack.org/82/639282/16/check/nova-tox-functional/a28bf42/ : SUCCESS in 21m 45s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639282/16/check/nova-tox-functional-py35/b226ed7/ : SUCCESS in 18m 34s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/16/check/tempest-slow-py3/b7d4f1e/ : SUCCESS in 2h 16m 57s","accounts_in_message":[],"_revision_number":16},{"id":"5bde5b1309b22a356f17873c0f7b15d63b2e79e2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 09:53:53.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/16/check/EMC_VxFlexOS_NOVA/3492b15/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":16},{"id":"eeaafc3b997356c5a97b9bfe4f195b189addf2a6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-12 10:41:18.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/16 : FAILURE in 17m 37s","accounts_in_message":[],"_revision_number":16},{"id":"b070018ca68bd5e0f1fd6cab48104340bc723c64","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 13:27:08.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"8735db6cfe28675b361be29a534d5f68464195cc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-12 15:17:23.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30432 : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":16},{"id":"cbedcb3cc00714f5665eb63744e60773cd03a11c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-12 22:21:13.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/16/check/tempest-dsvm-full-xenial/8367b2d/ : SUCCESS in 1h 57m 23s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/16/check/tempest-dsvm-full-xenial-py3/1e08a85/ : FAILURE in 2h 32m 02s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/16/check/grenade-dsvm-xenial/f5c776e/ : FAILURE in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"db49a556a3e51495956b01dabc0dc3e54da03c8f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-13 16:07:13.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"8f27a5a0d72e0d35db0efdcf4f1945aeb69159eb","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-16 17:44:24.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/16/check/check-dsvm-tempest-vz7-exe-minimal/4e64426 : SUCCESS in 1h 42m 53s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"66089dbd48043d89631710f1c5b324f3b8c77a6f","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-24 01:44:56.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"011a0625b15f7628f5a4f574abcf90a8c9d611d9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 01:58:29.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"92322937f99ba288b2cc222d0fae07e00ced97ab","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:05:53.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/17/check/check-dsvm-tempest-vz7-exe-minimal/a6cb0b1 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"3dbdccab05a27a9af5eea7116fb638ced23ff519","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 02:12:34.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/17/check/tempest-dsvm-full-xenial/2338df3/ : FAILURE in 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/17/check/tempest-dsvm-full-xenial-py3/ea0f7cc/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/17/check/grenade-dsvm-xenial/859f065/ : FAILURE in 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e76ac790c9b3b286e84bd71e23973a8e9d7fafe3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 02:14:01.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30972 : FAILURE in 7s","accounts_in_message":[],"_revision_number":17},{"id":"a08294af0979fd432dad5327f5e57f2b2746596f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:19:52.000000000","message":"Patch Set 17:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/531ee6f : FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":17},{"id":"f993275bef3e66f768c7702aac74c2a4d07d9e00","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-24 05:03:59.000000000","message":"Patch Set 17:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/17/check/nova-out-of-tree-pvm/fd505ba : FAILURE in 1m 21s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/17/check/nova-in-tree-pvm/dc56a3f : FAILURE in 1m 15s","accounts_in_message":[],"_revision_number":17},{"id":"2b5ec19ee5c91cc669747df52aa2f7870ebace0a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-24 05:52:22.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/17 : SUCCESS in 2h 11m 53s","accounts_in_message":[],"_revision_number":17},{"id":"419ae951301aa2d92e240adf2c53fddb066f85f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 06:58:34.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/17/check/grenade-py3/659b617/ : SUCCESS in 57m 39s\n- tempest-full-py3 http://logs.openstack.org/82/639282/17/check/tempest-full-py3/dcd631a/ : SUCCESS in 1h 39m 00s\n- openstack-tox-cover http://logs.openstack.org/82/639282/17/check/openstack-tox-cover/e93d4f0/cover/ : SUCCESS in 17m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/17/check/openstack-tox-lower-constraints/e4e7092/ : SUCCESS in 14m 41s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/17/check/openstack-tox-pep8/93e6ca0/ : SUCCESS in 9m 34s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/17/check/openstack-tox-py27/8578eba/ : SUCCESS in 12m 31s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/17/check/openstack-tox-py36/83aa4b1/ : FAILURE in 17m 32s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/17/check/openstack-tox-py37/2d6d75b/ : SUCCESS in 11m 49s\n- openstack-tox-docs http://logs.openstack.org/82/639282/17/check/openstack-tox-docs/1ea4f60/html/ : SUCCESS in 6m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/17/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/d1f4dec/ : SUCCESS in 52m 23s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/17/check/devstack-plugin-ceph-tempest/a89dfa5/ : SUCCESS in 1h 52m 08s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/17/check/neutron-grenade-multinode/e754559/ : SUCCESS in 1h 01m 24s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/17/check/neutron-tempest-linuxbridge/5285295/ : SUCCESS in 1h 43m 46s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/17/check/nova-grenade-live-migration/9658a3f/ : SUCCESS in 1h 08m 40s\n- nova-live-migration http://logs.openstack.org/82/639282/17/check/nova-live-migration/8310b88/ : SUCCESS in 50m 24s\n- nova-lvm http://logs.openstack.org/82/639282/17/check/nova-lvm/667b0b9/ : SUCCESS in 1h 07m 27s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/17/check/nova-next/99ec300/ : SUCCESS in 1h 40m 06s\n- nova-tox-functional http://logs.openstack.org/82/639282/17/check/nova-tox-functional/12b72f2/ : SUCCESS in 19m 47s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639282/17/check/nova-tox-functional-py36/b8b9f23/ : SUCCESS in 17m 37s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/17/check/tempest-slow-py3/5286ff3/ : SUCCESS in 2h 17m 41s","accounts_in_message":[],"_revision_number":17},{"id":"9d4513ff1b206454cdc50cce530920fcf1c5ecc9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 07:17:50.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/17/check-vote/ext-nova-zuul/2f7fab4 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":17},{"id":"915da44cd63d22405fed47a59cb233086120940e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:30:51.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"450653515bfdd3a2d1822accb999da17016627ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:51:36.000000000","message":"Uploaded patch set 20: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"3dfa2d9c0867c0ca959774cd4addf66949cb8817","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-25 19:46:20.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/20/check/tempest-dsvm-full-xenial/83e98d2/ : SUCCESS in 1h 14m 01s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/20/check/tempest-dsvm-full-xenial-py3/c4120ea/ : SUCCESS in 1h 27m 44s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/20/check/grenade-dsvm-xenial/82d432f/ : SUCCESS in 1h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"57da406087b77b01bfcca1d8c1d26526fe544306","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-25 21:59:51.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/20/check-vote/ext-nova-zuul/11362c7 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":20},{"id":"6787148bdec49874eec1a0b4ce0d2a4ef3eff035","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 00:09:16.000000000","message":"Patch Set 20:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33157 : FAILURE in 2h 43m 36s","accounts_in_message":[],"_revision_number":20},{"id":"0bcca486364d9b4c4f5cc742f6c02e1dbb2a123e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 00:22:17.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/20/check/check-dsvm-tempest-vz7-exe-minimal/7525b28 : FAILURE in 11m 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"5c65677324e64810c4c8bfa7da38738398b41314","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 01:07:10.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/20 : SUCCESS in 2h 24m 34s","accounts_in_message":[],"_revision_number":20},{"id":"cd1e5847a5dd4baf964bf04aceaff0bcd530f4d7","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 04:40:32.000000000","message":"Patch Set 20:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/20/check/nova-out-of-tree-pvm/1647f57 : FAILURE in 5h 05m 51s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/20/check/nova-in-tree-pvm/0199fe6 : FAILURE in 3h 13m 12s","accounts_in_message":[],"_revision_number":20},{"id":"e55163720d0a6292cf49d29a649c7bb31b59159b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 07:18:39.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/82/639282/20/check/grenade-py3/ae543f5/ : SUCCESS in 1h 05m 20s\n- tempest-full-py3 http://logs.openstack.org/82/639282/20/check/tempest-full-py3/1d85f90/ : SUCCESS in 1h 42m 07s\n- openstack-tox-cover http://logs.openstack.org/82/639282/20/check/openstack-tox-cover/78f4ac5/ : FAILURE in 13m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/20/check/openstack-tox-lower-constraints/646935a/ : FAILURE in 13m 47s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/20/check/openstack-tox-pep8/ef65d31/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/20/check/openstack-tox-py27/8c76d8e/ : FAILURE in 17m 01s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/20/check/openstack-tox-py36/082e962/ : FAILURE in 17m 07s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/20/check/openstack-tox-py37/11ce7fc/ : FAILURE in 10m 51s\n- openstack-tox-docs http://logs.openstack.org/82/639282/20/check/openstack-tox-docs/f5da995/html/ : SUCCESS in 10m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/20/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5086c5c/ : FAILURE in 36m 53s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/20/check/devstack-plugin-ceph-tempest/d30e965/ : SUCCESS in 1h 27m 58s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/20/check/neutron-grenade-multinode/0b56126/ : SUCCESS in 1h 07m 08s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/20/check/neutron-tempest-linuxbridge/c633e7d/ : SUCCESS in 2h 00m 47s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/20/check/nova-grenade-live-migration/17bf5ab/ : SUCCESS in 1h 20m 01s\n- nova-live-migration http://logs.openstack.org/82/639282/20/check/nova-live-migration/1b6e4d7/ : SUCCESS in 48m 46s\n- nova-lvm http://logs.openstack.org/82/639282/20/check/nova-lvm/f119374/ : SUCCESS in 1h 03m 13s (non-voting)\n- nova-multi-cell http://logs.openstack.org/82/639282/20/check/nova-multi-cell/888962a/ : SUCCESS in 1h 32m 22s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/20/check/nova-next/912c39a/ : SUCCESS in 1h 25m 21s\n- nova-tox-functional http://logs.openstack.org/82/639282/20/check/nova-tox-functional/b6bf781/ : SUCCESS in 21m 38s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639282/20/check/nova-tox-functional-py36/f9c28b0/ : SUCCESS in 18m 04s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/20/check/tempest-slow-py3/fd906ff/ : SUCCESS in 2h 42m 09s","accounts_in_message":[],"_revision_number":20},{"id":"870036b4992cc024f22c3a3bed0a2026f035d637","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-26 14:32:23.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"3622125b06803ae4927607be2ff2cc914fc608c0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 16:28:20.000000000","message":"Patch Set 21:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33198 : FAILURE in 1h 17m 59s","accounts_in_message":[],"_revision_number":21},{"id":"ecbb3022e5981eef975b69aa195b6e80da18e56d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-26 16:32:45.000000000","message":"Patch Set 21:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/21/check/tempest-dsvm-full-xenial/4cb31eb/ : FAILURE in 1h 21m 11s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/21/check/tempest-dsvm-full-xenial-py3/c3933a0/ : SUCCESS in 1h 34m 54s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/82/639282/21/check/grenade-dsvm-xenial/b8467c1/ : SUCCESS in 1h 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"caf9ef7fa3de7db35ec4f963d3107a188115bf16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-26 16:53:27.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/21/check-vote/ext-nova-zuul/79d6b8d : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":21},{"id":"16224b5caa61801ce05da540d30c524cf9ee4a53","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 18:17:11.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/21/check/check-dsvm-tempest-vz7-exe-minimal/c4f8b89 : FAILURE in 9m 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"a94b65b2648ec7a48858d16ee4c3bbc859122ed3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 19:28:53.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/21 : SUCCESS in 2h 28m 37s","accounts_in_message":[],"_revision_number":21},{"id":"16a6fdcc396003dd6cd633bc7a8f2053fb55b21a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 21:03:39.000000000","message":"Patch Set 21:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639282/21/check/nova-out-of-tree-pvm/9276225 : FAILURE in 5h 06m 34s\n- nova-in-tree-pvm http://184.172.12.213/82/639282/21/check/nova-in-tree-pvm/cd0aa6f : FAILURE in 5h 01m 29s","accounts_in_message":[],"_revision_number":21},{"id":"3d3d87ec3ae09f355383359f169642ea50ff9034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:19:55.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/21/check/grenade-py3/7a1a711/ : SUCCESS in 1h 07m 03s\n- tempest-full-py3 http://logs.openstack.org/82/639282/21/check/tempest-full-py3/f6e5632/ : SUCCESS in 1h 31m 18s\n- openstack-tox-cover http://logs.openstack.org/82/639282/21/check/openstack-tox-cover/1f57aa9/cover/ : SUCCESS in 13m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/21/check/openstack-tox-lower-constraints/94b75ec/ : SUCCESS in 13m 49s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/21/check/openstack-tox-pep8/7422c16/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/21/check/openstack-tox-py27/03e10e7/ : SUCCESS in 11m 30s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/21/check/openstack-tox-py36/1ea1b50/ : SUCCESS in 13m 53s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/21/check/openstack-tox-py37/2cad73a/ : SUCCESS in 11m 48s\n- openstack-tox-docs http://logs.openstack.org/82/639282/21/check/openstack-tox-docs/0784c52/html/ : SUCCESS in 6m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/21/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/06d5d6f/ : SUCCESS in 55m 33s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/21/check/devstack-plugin-ceph-tempest/92c2dcb/ : SUCCESS in 1h 58m 25s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/21/check/neutron-grenade-multinode/d1aedaf/ : SUCCESS in 1h 15m 20s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/21/check/neutron-tempest-linuxbridge/0c7002e/ : SUCCESS in 1h 29m 03s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/21/check/nova-grenade-live-migration/82a1c97/ : SUCCESS in 1h 12m 08s\n- nova-live-migration http://logs.openstack.org/82/639282/21/check/nova-live-migration/99c18ac/ : SUCCESS in 48m 14s\n- nova-lvm http://logs.openstack.org/82/639282/21/check/nova-lvm/654bd26/ : SUCCESS in 1h 04m 20s (non-voting)\n- nova-multi-cell http://logs.openstack.org/82/639282/21/check/nova-multi-cell/42246ec/ : SUCCESS in 1h 30m 55s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/21/check/nova-next/d68908b/ : SUCCESS in 1h 15m 21s\n- nova-tox-functional http://logs.openstack.org/82/639282/21/check/nova-tox-functional/e288219/ : SUCCESS in 19m 05s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639282/21/check/nova-tox-functional-py36/5ebb947/ : SUCCESS in 18m 37s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/21/check/tempest-slow-py3/4afca2f/ : SUCCESS in 2h 56m 02s","accounts_in_message":[],"_revision_number":21},{"id":"de1febdc49aee50da104175baf09287ed5d776de","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 03:21:33.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/21/check/EMC_VxFlexOS_NOVA/da96355/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":21},{"id":"26ee6578824b9bb02cddcefc7d1cbc35a0709859","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-27 07:04:51.000000000","message":"Patch Set 21:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-21 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":21},{"id":"a5d4573c1626715630259460441cc519812f263c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 13:49:29.000000000","message":"Patch Set 21:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/21/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/88449bd : SUCCESS in 2h 16m 11s","accounts_in_message":[],"_revision_number":21},{"id":"c890d0157f7c629af0ca72ad61c1acb0a887b3f7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-31 22:13:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"506968419c3d4b72d2ca047eb71f736b37303cd9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 22:19:38.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a611b817c2bce0a4e55c0e3cd989dfd501b416da","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-31 23:49:27.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/22/check/tempest-dsvm-full-xenial/43f5e48/ : SUCCESS in 1h 30m 08s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/22/check/tempest-dsvm-full-xenial-py3/63a86c5/ : SUCCESS in 1h 28m 05s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/22/check/grenade-dsvm-xenial/6c693c9/ : SUCCESS in 1h 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"02c9e44b92c73ff7af0fa3c472d020fdf6843978","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-07-31 23:57:29.000000000","message":"Patch Set 22:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-34549 : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":22},{"id":"d3ce38671548a59d166e9fda2d4e981f16ac303f","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 00:20:17.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/22/check/check-dsvm-tempest-vz7-exe-minimal/cf71f4c : FAILURE in 10m 21s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"4d4a63a46909befd0a7d17c6d8421cbeb743281b","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-01 00:50:50.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/639282/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"8ae429107832a8cd61a76e2c3f4a2ce9778d5135","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:27:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/82/639282/22/check/grenade-py3/043a63e/ : SUCCESS in 1h 05m 46s\n- tempest-integrated-compute http://logs.openstack.org/82/639282/22/check/tempest-integrated-compute/6d1d248/ : SUCCESS in 1h 40m 49s\n- openstack-tox-cover http://logs.openstack.org/82/639282/22/check/openstack-tox-cover/b1257ae/cover/ : SUCCESS in 22m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639282/22/check/openstack-tox-lower-constraints/aa66b0e/ : SUCCESS in 18m 43s\n- openstack-tox-pep8 http://logs.openstack.org/82/639282/22/check/openstack-tox-pep8/e1340c5/ : SUCCESS in 10m 15s\n- openstack-tox-py27 http://logs.openstack.org/82/639282/22/check/openstack-tox-py27/83524f5/ : SUCCESS in 14m 40s\n- openstack-tox-py36 http://logs.openstack.org/82/639282/22/check/openstack-tox-py36/7f878da/ : SUCCESS in 19m 32s\n- openstack-tox-py37 http://logs.openstack.org/82/639282/22/check/openstack-tox-py37/045048e/ : SUCCESS in 19m 49s\n- openstack-tox-docs http://logs.openstack.org/82/639282/22/check/openstack-tox-docs/aacb3ae/html/ : SUCCESS in 6m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/82/639282/22/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b7854ad/ : SUCCESS in 53m 55s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/82/639282/22/check/devstack-plugin-ceph-tempest/aa517e5/ : SUCCESS in 1h 09m 03s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/82/639282/22/check/neutron-grenade-multinode/8b8779f/ : SUCCESS in 1h 14m 11s\n- neutron-tempest-linuxbridge http://logs.openstack.org/82/639282/22/check/neutron-tempest-linuxbridge/d3a7b70/ : SUCCESS in 1h 26m 51s\n- nova-grenade-live-migration http://logs.openstack.org/82/639282/22/check/nova-grenade-live-migration/d81ff74/ : SUCCESS in 1h 13m 04s\n- nova-live-migration http://logs.openstack.org/82/639282/22/check/nova-live-migration/6fc25d3/ : SUCCESS in 47m 59s\n- nova-lvm http://logs.openstack.org/82/639282/22/check/nova-lvm/d311cc0/ : SUCCESS in 1h 00m 29s (non-voting)\n- nova-multi-cell http://logs.openstack.org/82/639282/22/check/nova-multi-cell/1bb806d/ : SUCCESS in 1h 35m 43s (non-voting)\n- nova-next http://logs.openstack.org/82/639282/22/check/nova-next/7f31029/ : SUCCESS in 1h 19m 46s\n- nova-tox-functional http://logs.openstack.org/82/639282/22/check/nova-tox-functional/6452fc2/ : SUCCESS in 23m 59s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639282/22/check/nova-tox-functional-py36/8df6096/ : SUCCESS in 22m 13s\n- tempest-slow-py3 http://logs.openstack.org/82/639282/22/check/tempest-slow-py3/2390449/ : SUCCESS in 2h 21m 40s\n- tempest-ipv6-only http://logs.openstack.org/82/639282/22/check/tempest-ipv6-only/6c05a92/ : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":22},{"id":"6f49d4d3c5ea985c568087c85279cf88854c510f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-01 02:20:19.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/82/639282/22/check-vote/ext-nova-zuul/0ce8c73 : FAILURE in 1h 01m 41s","accounts_in_message":[],"_revision_number":22},{"id":"760db93476dfaf8449278e987012f78b4c75d017","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-01 03:00:51.000000000","message":"Patch Set 22:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-22 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":22},{"id":"39f67f05bb837f90de4a108c0bb7da2a971aa447","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-01 07:03:19.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/82/639282/22/check-nova/Nova-ML2-Sriov/3f1c6f8 : SUCCESS in 1h 19m 48s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"f62f6b0869af318e3987f169f0c4bac8eb90abe3","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-06 07:00:18.000000000","message":"Patch Set 22: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"6b7e2f3a1cb5fb88441f5ba6e1b11089900eb3ff","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:26:20.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"f67bbfc1f7fe640307fd4bb11e5644f2d9c9cd43","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:27:37.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"ad19e7c3577508e5a78157432f6151b568f76933","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:31:01.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"698073b6fb200d6751827bae67e1842ed8150d31","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:31:45.000000000","message":"Patch Set 23:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/23/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/2c45971 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":23},{"id":"92ac68135f5482fc51cb31a26f51ab3f5647bf7a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:40:33.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"301a099195ef9e03523bd0ac4c35521d9b02d41f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-16 11:40:02.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/639282/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"94fcfc29f04e3939cf69fa75c89b71938873c3b9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-16 12:24:01.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35277 : FAILURE in 1h 48m 49s","accounts_in_message":[],"_revision_number":23},{"id":"1ce8bee895e88bdef0251cf3c709ce33885cf3e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:01:10.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac58bc3d394f4c6ea33c12a2379a0938 : SUCCESS in 1h 23m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dfac4bff1f5d4703aeb9f47d13c13775 : SUCCESS in 1h 16m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfcd831044c44ef3a463ce00ff804b0f : SUCCESS in 19m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b4e56ccb07240c693f3f04d23bb77f9 : SUCCESS in 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e7f9c5edc0e4ccf81ba1f55ed246166 : SUCCESS in 13m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdeae6900e03461dbde80a96048ba127 : SUCCESS in 18m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae308c03f72244a2b0bedaa9697ae63f : SUCCESS in 16m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2dc42753e85406385f3ddf3545d1e43 : SUCCESS in 19m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5ee273191e46178ae6aaec1f0eab0d : SUCCESS in 8m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/285d4265253441e1bed4738c4e2acf15 : SUCCESS in 57m 35s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f90e69254f91407c80aa7bf4121e5fc8 : SUCCESS in 1h 09m 50s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a752aaf84f14408a8e83e384f68e099 : SUCCESS in 1h 16m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6b9df8f004c849019eae5f1e51b06a65 : SUCCESS in 1h 26m 16s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/f2b4d59566ad437aab7e8b73b0ca586b : SUCCESS in 1h 11m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7da3a9e86922422492ce61c97ea3bea0 : FAILURE in 1h 06m 29s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a6c69ebab5034be98a3d8364aeb1e565 : SUCCESS in 49m 36s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b5c6164943a242f89e99d7b2e8f4540f : SUCCESS in 1h 17m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/e5a30a442f344480aedf4d58a659cb8e : SUCCESS in 1h 16m 43s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/794edd164c7c445c9e728e53b3368e9a : SUCCESS in 20m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7b3da2424584ff8b51c314347fb1e3f : SUCCESS in 17m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3219e3021944d1481bda1fc01f871e0 : SUCCESS in 2h 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a735ba7744f440b8834c909c76499dbb : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":23},{"id":"4dab988e2c8467d3f7e3f15bf2c26e6322a64896","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-16 13:30:43.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"10eda1cfb99829554f394437dc3017c0cb0d6f9b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 13:32:05.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"aa8dcb57aab36b8b21d1c0c62b00412e3422f1fd","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-16 14:28:46.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35287 : FAILURE in 57m 00s","accounts_in_message":[],"_revision_number":23},{"id":"27e0c1848a443833f862c9fbff8006f74144a56d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 16:10:20.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be2c2fe5d64f4692bda06326527960d3 : SUCCESS in 1h 10m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0d0752cf303340faa5155d19a5dbeba8 : SUCCESS in 1h 41m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3bc3cf5d1cf4ba09b6914c885441b75 : SUCCESS in 19m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/684a386d359340c787653818c26f7376 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f81cec6ad547c3aff5fa7981108c35 : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b2e9405c0ed42eb8d58176f5fa6bb95 : SUCCESS in 22m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/996b1c94e1264276bfa4d8908d07c2c5 : SUCCESS in 18m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abaa39dd9e90433db5569d3ac3c50f2b : SUCCESS in 19m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2255d5f9e1ff4d20b0209aef67372173 : SUCCESS in 8m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/265e03e99e9f4441b1358f92d717e22a : SUCCESS in 1h 00m 01s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6837f57872b4876bf5348509e65984e : SUCCESS in 1h 32m 49s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d49644eaf3a46d29c2a708f946ffeb3 : SUCCESS in 1h 12m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/44aec34243d04280bba75721b16aa434 : SUCCESS in 1h 34m 33s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/71d2b5f5b3d74ecf8e82bb9746471f27 : SUCCESS in 1h 18m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/582c2c56daca4a7292ad61f940877556 : SUCCESS in 57m 03s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/370c9aef91bc493ba3450eed9668738b : SUCCESS in 57m 15s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/af29881f23ac4e1a9bea50098f49975d : SUCCESS in 1h 24m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/56768edf22a846f89515f7fda15fa30d : SUCCESS in 1h 27m 57s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/28869d8858b3464a81562d5a913ac508 : SUCCESS in 18m 35s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47703867a75146b2bf6f5b35506b51d2 : SUCCESS in 17m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c7e8e713b2343d8a64d48dc968dfb83 : SUCCESS in 2h 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2555f6f1226d4405864c89e966742d96 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":23},{"id":"fb6ce2c465658e519184498102df59bc931a37d1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 16:11:10.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/23/check/EMC_VxFlexOS_NOVA/e6a3661/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"fe160627785c53e115a6b1f3556941edcd6b2a3b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-16 18:15:46.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/23 : SUCCESS in 2h 20m 24s","accounts_in_message":[],"_revision_number":23},{"id":"0c841f6bc50acbfd105be6f949e03ee4b3054ec1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-17 01:16:21.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"117b142d54670c7dfa348f7fa6de0aca7b44c3ab","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 02:24:55.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/23/check/check-dsvm-tempest-vz7-exe-minimal/66d9641 : FAILURE in 10m 41s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"cf35afe455124ee7c29a0a0e1f6090f343ce13b9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-19 12:48:39.000000000","message":"Patch Set 23: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":23},{"id":"3d02c871393d8e64184130c947f56809a412939b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 16:42:53.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"b1fb4286fc974a67852612b5ce8bd747b9b60558","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-20 11:14:01.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"72d2d8911cf53fb744acfe89a0e72ec1b350b828","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-20 11:34:45.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"c859bfcc3ba303496218ba55414ffe717248d3b7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 11:35:33.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"586cf4f19326e88d6d3317f4ad4583c0cc6c4a06","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 13:01:14.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/82/639282/24/check/check-dsvm-tempest-vz7-exe-minimal/a97b7aa : FAILURE in 10m 29s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"83321586318ee3042fb1e7a9e5face2547e24126","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 13:22:41.000000000","message":"Patch Set 24:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639282/24/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/13f9e5c : FAILURE in 44m 45s","accounts_in_message":[],"_revision_number":24},{"id":"bbe3b93dca9142042bd0971af2a51438f0c97787","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-20 13:33:06.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35417 : FAILURE in 1h 57m 29s","accounts_in_message":[],"_revision_number":24},{"id":"d0604def85693b3b7534909bbd4ef5f9b1e89273","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-20 13:56:03.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/tempest-dsvm-full-xenial/3366d07/ : SUCCESS in 1h 30m 35s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/tempest-dsvm-full-xenial-py3/47bc74a/ : SUCCESS in 1h 34m 52s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/grenade-dsvm-xenial/3a752bf/ : SUCCESS in 2h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d3ea33759c1b17613147058fc3a707c00c4c077c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:12:31.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09c2d4d8f22647959ace043e051d595a : SUCCESS in 1h 05m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a4314b72f5014e8297f7b81e95efa6af : SUCCESS in 1h 30m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa4ac4522d6a435e84403038522d42d2 : SUCCESS in 18m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eb69aa1b0814bf6a3718fad4a90b848 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d153790f374dc489cdf52b0cc71fd2 : SUCCESS in 12m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/866d1f74627f493f87b6b3224743507d : SUCCESS in 14m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb7dcc1b20bc4cc58ea29f4c02719f7e : SUCCESS in 13m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e6a269e774543288625e6df91713abb : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e903e7765ce341f4b1f549bcd156cae3 : SUCCESS in 7m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/155a05b73caf4df2867c5e240feb32dd : SUCCESS in 52m 58s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/475e54cf2bf648cdb9dbbd15064480fc : SUCCESS in 1h 20m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05e0965cb47d41aab3a7c84e3b526884 : SUCCESS in 1h 16m 55s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5d446e7edeaa41d0a933575e33e9cee7 : SUCCESS in 1h 18m 29s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/0fe0006ecf584d0eb7d1367345335778 : SUCCESS in 1h 13m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0575194d79ec4f45af9df3a77cbc19c0 : SUCCESS in 50m 38s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/414cf6eb94a9402ea3429fd4f4b8441e : SUCCESS in 49m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5cf92f86493740b79043a56ca52b73c8 : SUCCESS in 1h 35m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/b638ad16cc574103aac3e7bfd907aab9 : SUCCESS in 1h 31m 03s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/5b0d1420c2fa4fcaa2825c0e85ae1b81 : SUCCESS in 19m 18s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aa4a6566a1c44b6abc89124a1449431 : SUCCESS in 16m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9692af5da6fc484da2112aeeeee0b643 : SUCCESS in 2h 29m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeaae198325645ea8fddbe2d1543d798 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":24},{"id":"008a7a77e736010c58665226b4d048c85055b272","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 14:13:20.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/24/check/EMC_VxFlexOS_NOVA/1578a45/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":24},{"id":"f7d6b8ea60ade21dbe937e613ca0b3b5160670fd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-20 14:31:59.000000000","message":"Patch Set 24:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639282/24 : FAILURE in 2h 55m 43s","accounts_in_message":[],"_revision_number":24},{"id":"88e20761f5709298f87f096e1179bdc7a97c36bd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-20 15:52:42.000000000","message":"Patch Set 24:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-24 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":24},{"id":"be1d9cbcdebb4cffc53c4bfc63a4d316783965f5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-20 19:14:00.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"ab5444ed6d64ba1211caa79d4c55b9b2372cc59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:14:22.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"ce92f17477df1df5fbf4279c6368ad4f79fc1b35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 01:02:43.000000000","message":"Patch Set 24: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5b45ef1217964b82b5aa339563a029ce : SUCCESS in 1h 06m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ae03ca664f2b422ea53f3a9486429b1b : SUCCESS in 1h 27m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d86850fdaa1e4702a04c69926d681e27 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbb06c7bfb64f3db856a8d5f8c62feb : SUCCESS in 11m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a683808bbb84e588698345c08804794 : FAILURE in 31m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/122a05712f644c479f1f847da4742e9b : SUCCESS in 13m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75226c75ce054ff885f42424b9814e2f : SUCCESS in 17m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7dc3d06e7214f48966b8d8386084979 : SUCCESS in 8m 24s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/a6e0ef0efcc64b6f987c82357e26c755 : SUCCESS in 1h 23m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a05dffda64c742c0992f356fa0cd7e7b : SUCCESS in 57m 14s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/78fa61b6fb1242f2af0eb9a12c202d93 : SUCCESS in 20m 20s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d1ecc12f23a4aae841d6069aac345a2 : SUCCESS in 21m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a8b465d4d87d4c3988f52c06e63ebc59 : SUCCESS in 1h 40m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/c4c9eed28e6c452bb841fda6b48b5c38 : SUCCESS in 1h 43m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd8758c22f564e9f971497e7584f599d : SUCCESS in 2h 25m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0262d06f598490d8e6551f0f58d83ac : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":24},{"id":"c99551c6f54366321e34eaef3ca419e761b5ffbf","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-21 06:58:23.000000000","message":"Patch Set 24:\n\nrecheck bug 1793364","accounts_in_message":[],"_revision_number":24},{"id":"66904716aa01fba06104c7649df34d82bac431de","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 06:58:52.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2123122ef95c986d4842fbcb046faa3966b2f629","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 10:11:35.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/395266c69b0946cb94d02ffe62a7a244 : SUCCESS in 54m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/eccfa3230f784205b7808a73da7454ff : SUCCESS in 1h 12m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e178940037b149e384c9ccc8ba3e92c1 : SUCCESS in 18m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69e9de56ea534752b0060fe7154e1257 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21accd07b5af4677b69da46255371c62 : SUCCESS in 11m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c9526fbdced4361beca00c3d60c8491 : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12a7b26ab87248e3ab45a5e9fb0e2a39 : SUCCESS in 15m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/809e351a87334960a143584525d08db7 : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1981c8ba8c69404eb2c2c883a81b7b80 : SUCCESS in 7m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0dfe0b08009046d9a5616cef57e0b4d9 : SUCCESS in 54m 53s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/862c17ee26eb4acfa3bf7d19058035bd : SUCCESS in 1h 15m 25s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/015b7ed6d5d943a68fae81954161c28c : SUCCESS in 1h 08m 41s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/975c1c0e09a74e8da53feeb153d6dd20 : SUCCESS in 1h 24m 02s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/5ec31024e51d454eba0661ad82762728 : SUCCESS in 1h 11m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f4bfd9c8ecca4719945ef2d3dc86d18a : SUCCESS in 57m 15s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8af7b5ba95a24fda91de8f606296f29e : SUCCESS in 54m 34s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/58f8235338df42cb8716d909c992805b : SUCCESS in 1h 19m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/cc8c911516ec439a98c712bdecb72757 : SUCCESS in 1h 27m 52s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/c3e4a285f1c2449ebcad92b9d0396852 : SUCCESS in 24m 14s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8db04e9ad1740c98a523019ae7927b0 : SUCCESS in 17m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/383a7f367fb845a181409f35ae097f09 : SUCCESS in 2h 48m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ac9915c4a0e41a79725e7a403073df0 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":24},{"id":"6090b1bd1fc1ca6bc4fb1cfb41b4650cd1b4cf5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 10:12:08.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"2aa11615cf380aee7a6251b0d1773c603a20f913","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 10:12:11.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639282/24/check/EMC_VxFlexOS_NOVA/dd85827/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":24},{"id":"7bd5e83368510caa6d2523b2e787d7d13ab8cd24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-21 11:22:15.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/tempest-dsvm-full-xenial/25344bb/ : SUCCESS in 2h 23m 34s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/tempest-dsvm-full-xenial-py3/f3ceb91/ : SUCCESS in 2h 05m 25s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/82/639282/24/check/grenade-dsvm-xenial/1d145b6/ : SUCCESS in 1h 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"262fefd220408d8beee64462ed7d6a893ff319b3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-21 12:25:11.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639282/24 : SUCCESS in 2h 26m 57s","accounts_in_message":[],"_revision_number":24},{"id":"262fef6cc62a906e4fb9b965f9e04ce91859eec4","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-21 14:01:48.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/82/639282/24/check-nova/Nova-ML2-Sriov/7cf5d89 : SUCCESS in 1h 23m 10s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/82/639282/24/check-nova/Nova-MACVTAP-ML2-Sriov/a80df51 : FAILURE in 1h 59m 04s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":24},{"id":"6d4f9bbdbf9e9053a32ecfd26cbb4434c941c8af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:29:57.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/70edd26fa2f545f1b729e0aba7864982 : SUCCESS in 1h 12m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c3c4994a3a0b4e4caf01991b5498350d : SUCCESS in 1h 16m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/874cb2a4b4984651a5d7d75b22becec5 : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f56b08e57c42c78898f114524c8858 : SUCCESS in 10m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/065d84a14adb4c2d8daa275ff622ff27 : SUCCESS in 20m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43f2fcc31e304013b80c05949c0d43f7 : SUCCESS in 19m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fde78d71581749718ad22d75422f2ce8 : SUCCESS in 17m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9714a2faa8f74f55b59046dc93843c66 : SUCCESS in 10m 29s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/29f7f9573cbe4f60a0534c5ebcdc3896 : SUCCESS in 1h 10m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4d2b0755399e4f29bcfd30b9d5890911 : SUCCESS in 54m 35s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/9ba8e2a91315422eb8b71418741c5d37 : SUCCESS in 24m 19s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92a9a4141d784a86b289e24dd9c8b219 : SUCCESS in 23m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/443bf7b789bc4ed095a9e1b6e23945c9 : SUCCESS in 1h 31m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/f4e950806d594a4b9644bb5e5ff9404d : SUCCESS in 1h 22m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a3b87a5eaa264f63a97eea586783f3fb : SUCCESS in 2h 23m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf4ee8c8eec4cd9baac212e276a9018 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":24},{"id":"9cf470e6bae3505a229410fcd663bbba1f63d98e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:29:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24},{"id":"2ac839205c6f012ddd4cb3bcfd9dc3de57cfb618","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-21 17:31:14.000000000","message":"Patch Set 24:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639282-24 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":24},{"id":"cf20b4277e5bc31bd9a538bf300af3986f74b8d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:32:12.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc85db2d2a2428186572d8f657f34d3 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"2c29d54fb72c4d77dcfef7f00ffbc465057b18ff","revisions":{"3fb3fe3d8e138f8d329ba17b94446fea403b64a5":{"kind":"REWORK","_number":1,"created":"2019-02-26 09:43:48.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/1"}}},"commit":{"parents":[{"commit":"cd6c272d25dbfc8032f2b69a0a7f7a4690f6cfb7","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd6c272d25dbfc8032f2b69a0a7f7a4690f6cfb7"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 09:43:32.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fb3fe3d8e138f8d329ba17b94446fea403b64a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fb3fe3d8e138f8d329ba17b94446fea403b64a5"}]},"branch":"refs/heads/master"},"b8cb4e535a734a6e8ef4877f263d23d862bb891c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-03-05 22:35:50.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/2"}}},"commit":{"parents":[{"commit":"597d2c8827c39cc409cc3b01003cb8e66f8219e7","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/597d2c8827c39cc409cc3b01003cb8e66f8219e7"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-05 22:18:10.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8cb4e535a734a6e8ef4877f263d23d862bb891c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8cb4e535a734a6e8ef4877f263d23d862bb891c"}]},"branch":"refs/heads/master"},"ec10483663f8d208ba449fefa3261bf03410e2a2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-06 19:09:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/82/639282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/3"}}},"commit":{"parents":[{"commit":"359151cf49eff0eafcad0d7c52ca71c416a7a3d2","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/359151cf49eff0eafcad0d7c52ca71c416a7a3d2"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-03-06 19:51:05.000000000","tz":-480},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec10483663f8d208ba449fefa3261bf03410e2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ec10483663f8d208ba449fefa3261bf03410e2a2"}]},"branch":"refs/heads/master"},"b6ed046085d77a6b3195678bba2e9e29b5ffe9f4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-07 12:06:34.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/4"}}},"commit":{"parents":[{"commit":"1ba46dedc373eb96addc6673376ee276738a71c1","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ba46dedc373eb96addc6673376ee276738a71c1"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-07 12:05:58.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6ed046085d77a6b3195678bba2e9e29b5ffe9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6ed046085d77a6b3195678bba2e9e29b5ffe9f4"}]},"branch":"refs/heads/master"},"cd51d9ffb6a73dd9799077add0e2c6536371cace":{"kind":"REWORK","_number":5,"created":"2019-03-08 05:17:40.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/5"}}},"commit":{"parents":[{"commit":"23242ca8197d6fada0329fa0f5250435b96b57f0","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23242ca8197d6fada0329fa0f5250435b96b57f0"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-08 00:12:30.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd51d9ffb6a73dd9799077add0e2c6536371cace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd51d9ffb6a73dd9799077add0e2c6536371cace"}]},"branch":"refs/heads/master"},"ea780bd28195a2ad26fef3b883ee91b8fdc5b08e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-18 05:44:49.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/6"}}},"commit":{"parents":[{"commit":"30d245379753706d3d57348b5f4eaf907f30ab24","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30d245379753706d3d57348b5f4eaf907f30ab24"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-17 23:19:09.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea780bd28195a2ad26fef3b883ee91b8fdc5b08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea780bd28195a2ad26fef3b883ee91b8fdc5b08e"}]},"branch":"refs/heads/master"},"48715c481da361a5ad96406e060f74bab8c911f8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-21 09:37:18.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/7"}}},"commit":{"parents":[{"commit":"24e64e7943f9d7fb9331d6daba3e17340eb8a1d4","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24e64e7943f9d7fb9331d6daba3e17340eb8a1d4"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-21 09:37:01.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48715c481da361a5ad96406e060f74bab8c911f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48715c481da361a5ad96406e060f74bab8c911f8"}]},"branch":"refs/heads/master"},"79fed930f45228c799fd65af18c27f8005a6c30f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-03-27 04:38:51.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/8"}}},"commit":{"parents":[{"commit":"982657c0286813954b54a2b01727e9845f24b880","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/982657c0286813954b54a2b01727e9845f24b880"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-26 23:16:52.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79fed930f45228c799fd65af18c27f8005a6c30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79fed930f45228c799fd65af18c27f8005a6c30f"}]},"branch":"refs/heads/master"},"ad615d4b8a58b369e8abe397333dddd157726b48":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-03-29 04:13:05.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/9"}}},"commit":{"parents":[{"commit":"319b329a4b1d29e838d83daf17ca8f2978483132","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/319b329a4b1d29e838d83daf17ca8f2978483132"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 04:10:28.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad615d4b8a58b369e8abe397333dddd157726b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad615d4b8a58b369e8abe397333dddd157726b48"}]},"branch":"refs/heads/master"},"95402e646dd677aa46274d4396a58724447075e8":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-29 06:09:26.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/10"}}},"commit":{"parents":[{"commit":"c346d19ec2605072619e4f6613648fce0d85dc1a","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c346d19ec2605072619e4f6613648fce0d85dc1a"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 06:04:21.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95402e646dd677aa46274d4396a58724447075e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95402e646dd677aa46274d4396a58724447075e8"}]},"branch":"refs/heads/master"},"43eb74935a456d7a8089fcc0683f722a8c970636":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-03-29 20:45:30.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/82/639282/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/11"}}},"commit":{"parents":[{"commit":"903547c5b072ceac4201ba218c9052e698eaa78b","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/903547c5b072ceac4201ba218c9052e698eaa78b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-03-29 20:41:16.000000000","tz":-300},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43eb74935a456d7a8089fcc0683f722a8c970636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43eb74935a456d7a8089fcc0683f722a8c970636"}]},"branch":"refs/heads/master"},"4e12383281de3945c79c9bce066a605c2c817a88":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-01 21:18:06.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/12"}}},"commit":{"parents":[{"commit":"2bfce7c82388ab5e50540b424b038b3abc533b2f","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bfce7c82388ab5e50540b424b038b3abc533b2f"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-01 21:15:55.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e12383281de3945c79c9bce066a605c2c817a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e12383281de3945c79c9bce066a605c2c817a88"}]},"branch":"refs/heads/master"},"90d3556c8ab5145e578f3877b55f15b989299931":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-04-02 10:13:10.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/13"}}},"commit":{"parents":[{"commit":"44839932b32c1fff6a39b8e455186a35de24e4de","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44839932b32c1fff6a39b8e455186a35de24e4de"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-02 10:11:05.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90d3556c8ab5145e578f3877b55f15b989299931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90d3556c8ab5145e578f3877b55f15b989299931"}]},"branch":"refs/heads/master"},"28d4aa33a765f152764b5182bf0463312e9c4dac":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-04 08:44:44.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/14"}}},"commit":{"parents":[{"commit":"882987b4c36d8e0140360252a72daec327262bb5","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/882987b4c36d8e0140360252a72daec327262bb5"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 08:43:47.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28d4aa33a765f152764b5182bf0463312e9c4dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28d4aa33a765f152764b5182bf0463312e9c4dac"}]},"branch":"refs/heads/master"},"561ee1dd34494d7a21e65e215f56794ee51f1a7e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-04-04 20:44:48.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/15"}}},"commit":{"parents":[{"commit":"a66bb1faa6e394af4c09c68ba65f5b749e968f84","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a66bb1faa6e394af4c09c68ba65f5b749e968f84"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 20:44:03.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/561ee1dd34494d7a21e65e215f56794ee51f1a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/561ee1dd34494d7a21e65e215f56794ee51f1a7e"}]},"branch":"refs/heads/master"},"301573d1c7b549f4ff4e15cb94b4eaf2863e546d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-12 06:57:52.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/16"}}},"commit":{"parents":[{"commit":"3153d05c6fddb7c2423d608ee2a57bc30c690aa8","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3153d05c6fddb7c2423d608ee2a57bc30c690aa8"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-11 22:22:09.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/301573d1c7b549f4ff4e15cb94b4eaf2863e546d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/301573d1c7b549f4ff4e15cb94b4eaf2863e546d"}]},"branch":"refs/heads/master"},"bf81fd244d912476f80f8db3274d30713a30dbc8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-04-24 01:44:56.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/82/639282/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/17"}}},"commit":{"parents":[{"commit":"17017b1816a91af89700f368c2b67a3474146e70","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17017b1816a91af89700f368c2b67a3474146e70"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-23 22:12:31.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf81fd244d912476f80f8db3274d30713a30dbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf81fd244d912476f80f8db3274d30713a30dbc8"}]},"branch":"refs/heads/master"},"9b12a2e0da1a0034bc48ad6f3e3586c1575da6a7":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-06-25 16:30:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/639282/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/18"}}},"commit":{"parents":[{"commit":"ca4f99ddaabd438433ef5830d681a3472e154e95","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca4f99ddaabd438433ef5830d681a3472e154e95"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:24:50.000000000","tz":60},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b12a2e0da1a0034bc48ad6f3e3586c1575da6a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b12a2e0da1a0034bc48ad6f3e3586c1575da6a7"}]},"branch":"refs/heads/master"},"b228c0d92dfa7231c08f1863f01ebcaf8d7a929d":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-06-25 16:51:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/639282/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/20"}}},"commit":{"parents":[{"commit":"94c7e0e04781711063a11e4b8e6607b2383cb557","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/94c7e0e04781711063a11e4b8e6607b2383cb557"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:51:20.000000000","tz":60},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b228c0d92dfa7231c08f1863f01ebcaf8d7a929d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b228c0d92dfa7231c08f1863f01ebcaf8d7a929d"}]},"branch":"refs/heads/master"},"035a94e4d56d3cd063667edb33915c1c00a20d4d":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-06-26 14:32:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/639282/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/21"}}},"commit":{"parents":[{"commit":"b82b22f182dca6ab074fd4fd373d70c78cfa2f0e","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b82b22f182dca6ab074fd4fd373d70c78cfa2f0e"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-26 14:30:20.000000000","tz":60},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/035a94e4d56d3cd063667edb33915c1c00a20d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/035a94e4d56d3cd063667edb33915c1c00a20d4d"}]},"branch":"refs/heads/master"},"82c61a80e4ce185493a742d0a82b1a6ec3e9307a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-07-31 22:13:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/82/639282/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/22"}}},"commit":{"parents":[{"commit":"4640d2b2022ed247830d2cd84fb368e8934a9362","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4640d2b2022ed247830d2cd84fb368e8934a9362"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-31 22:13:27.000000000","tz":-300},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82c61a80e4ce185493a742d0a82b1a6ec3e9307a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82c61a80e4ce185493a742d0a82b1a6ec3e9307a"}]},"branch":"refs/heads/master"},"536af89b11ea69f3a13c4692301432290a1d0690":{"kind":"REWORK","_number":23,"created":"2019-08-16 10:27:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/639282/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/23"}}},"commit":{"parents":[{"commit":"f2d93d09802bc3b9ae40e89088cce3fb34c47b52","subject":"Move router advertisement daemon restarts to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2d93d09802bc3b9ae40e89088cce3fb34c47b52"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-16 10:25:21.000000000","tz":60},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/536af89b11ea69f3a13c4692301432290a1d0690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/536af89b11ea69f3a13c4692301432290a1d0690"}]},"branch":"refs/heads/master"},"2c29d54fb72c4d77dcfef7f00ffbc465057b18ff":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-08-20 11:34:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/639282/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639282/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639282/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639282/24"}}},"commit":{"parents":[{"commit":"7e982537290e9d5eb260d7f71d789757e5c0d9be","subject":"Merge \"Move router advertisement daemon restarts to privsep.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e982537290e9d5eb260d7f71d789757e5c0d9be"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-26 03:07:24.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-02-26 03:07:24.000000000","tz":0},"subject":"Move calls to ovs-vsctl to privsep.","message":"Move calls to ovs-vsctl to privsep.\n\nThe exception handling isn\u0027t perfect, but it keeps the interface\nunchanged. Its quite hard to include the command in the exception\nlike we did before, so I am hoping we can just not do that thing.\n\nChange-Id: I9d46e0dc228ed2df9f681c1a2f4dca3c018c7822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c29d54fb72c4d77dcfef7f00ffbc465057b18ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c29d54fb72c4d77dcfef7f00ffbc465057b18ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
