)]}'
{"id":"openstack%2Fnova~639382","triplet_id":"openstack%2Fnova~master~If6a36d23768877bfa820ed44b610bfb113df5464","project":"openstack/nova","branch":"master","topic":"bp/cross-cell-resize","hashtags":[],"change_id":"If6a36d23768877bfa820ed44b610bfb113df5464","subject":"Improve CinderFixtureNewAttachFlow","status":"MERGED","created":"2019-02-26 16:40:42.000000000","updated":"2019-04-24 04:54:42.000000000","submitted":"2019-04-24 04:54:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639382-1556081682657-776347d6","meta_rev_id":"a14c32bc5639faf43071c0e0f1e883a9b592308d","_number":639382,"virtual_id_number":639382,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-20 23:09:33.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-20 22:28:13.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-20 22:45:17.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-04-20 22:44:58.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-04-20 22:15:58.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-04-21 03:59:35.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-04-24 04:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-20 22:13:39.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-23 13:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-23 13:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-03 21:35:54.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-11 20:21:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-20 22:15:58.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-20 22:28:13.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-20 22:44:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-20 22:45:17.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-04-20 23:09:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-21 03:59:35.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-04-23 13:27:30.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-04-24 04:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aba431b6501312c8701df8c709b33ff00189ea4f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-26 16:40:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc21079dbea95865eb15503fa728ccd2ff0887bf","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 16:54:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e275e7298c7e72df13360b0b6f49a357e2b7436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 17:23:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/1/check/openstack-tox-cover/7d5b2ac/cover/ : SUCCESS in 18m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/1/check/openstack-tox-lower-constraints/58fb76f/ : SUCCESS in 22m 57s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/1/check/openstack-tox-pep8/ee20ee8/ : SUCCESS in 9m 55s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/1/check/openstack-tox-py27/0e46aba/ : SUCCESS in 23m 30s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/1/check/openstack-tox-py35/c618802/ : SUCCESS in 12m 04s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/1/check/openstack-tox-py36/c37ddd0/ : SUCCESS in 12m 28s\n- openstack-tox-docs http://logs.openstack.org/82/639382/1/check/openstack-tox-docs/91bf604/html/ : SUCCESS in 6m 07s\n- nova-tox-functional http://logs.openstack.org/82/639382/1/check/nova-tox-functional/b3f197b/ : FAILURE in 20m 51s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/1/check/nova-tox-functional-py35/c4ae1fe/ : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d504b3efbdcc546260cd0a1a8ab64438a94a8bc0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-26 19:07:16.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28287 : FAILURE in 1h 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2cf999c7a6e66d79de79f8a26f3aee63634a8124","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-26 19:43:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/1 : SUCCESS in 2h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d37d65b390c01c9354cff1d5d152845725edfcee","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-26 22:16:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"76c7265f8f6c13101cab3bb2f0fc8f66899f9834","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 22:18:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ebe1ca821aa3b42f52849ec659910fc91748b77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 23:26:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/2/check/openstack-tox-cover/895a76d/cover/ : SUCCESS in 14m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/2/check/openstack-tox-lower-constraints/405a402/ : SUCCESS in 12m 51s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/2/check/openstack-tox-pep8/9c3c788/ : SUCCESS in 8m 26s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/2/check/openstack-tox-py27/63207b8/ : SUCCESS in 15m 35s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/2/check/openstack-tox-py35/fef9476/ : SUCCESS in 12m 39s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/2/check/openstack-tox-py36/6739e89/ : SUCCESS in 11m 29s\n- openstack-tox-docs http://logs.openstack.org/82/639382/2/check/openstack-tox-docs/e04027c/html/ : SUCCESS in 5m 58s\n- nova-tox-functional http://logs.openstack.org/82/639382/2/check/nova-tox-functional/1d01d99/ : SUCCESS in 17m 20s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/2/check/nova-tox-functional-py35/1c22313/ : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9db59251d1fcb5afd1207cc468779f64d059e8b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 23:27:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/2/check/EMC_VxFlexOS_NOVA/bac8233/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"06ddb93f9fc4788c93979c7ba20186501edc3bec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-27 00:12:17.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"d13c150c90504f7780e207dd23fabe065fae9d6f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-27 00:40:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/2 : SUCCESS in 2h 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"81e6a54f2c1debac64c71576072c6026bed30c26","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-27 03:15:51.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/2/check-vote/ext-nova-zuul/6b2bb01 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"26d2a441e8a03d38d79626d073c439013cc1aea2","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-07 22:59:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"804b42e9480cb0beccdab0219550ec5c395e26f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 23:59:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/3/check/openstack-tox-cover/67d4a9f/cover/ : SUCCESS in 17m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/3/check/openstack-tox-lower-constraints/bd00b61/ : SUCCESS in 14m 17s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/3/check/openstack-tox-pep8/4775f02/ : SUCCESS in 11m 26s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/3/check/openstack-tox-py27/e8a5acc/ : SUCCESS in 12m 59s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/3/check/openstack-tox-py35/62cd307/ : SUCCESS in 13m 33s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/3/check/openstack-tox-py36/c3761a1/ : SUCCESS in 13m 44s\n- openstack-tox-docs http://logs.openstack.org/82/639382/3/check/openstack-tox-docs/afcbd95/html/ : SUCCESS in 6m 57s\n- nova-tox-functional http://logs.openstack.org/82/639382/3/check/nova-tox-functional/712e060/ : SUCCESS in 21m 45s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/3/check/nova-tox-functional-py35/3b584fd/ : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":3},{"id":"efbbadb12e903cb5e5353bb3e13eb373c7540bcd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 00:00:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/3/check/EMC_VxFlexOS_NOVA/c5079d7/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"b779263232798c16b753fbea07d9f57741f855eb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 02:07:33.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/3 : FAILURE in 2h 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"aa4bcbd7e0176c83501d4fd5ed04a7e3fce368a9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-08 05:15:22.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"ac9373cbb46bb6d265ce5f724c1a790de2e2264a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-08 09:06:38.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/3/check-vote/ext-nova-zuul/529245f : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7f076bc610e8be9775cf331b7026545e22a2fa78","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-14 20:34:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e06934152960d1db11a643d93c51a7ed30927386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-14 22:08:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/4/check/openstack-tox-cover/17e1bb8/cover/ : SUCCESS in 15m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/4/check/openstack-tox-lower-constraints/731d79d/ : SUCCESS in 11m 02s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/4/check/openstack-tox-pep8/c0991b0/ : SUCCESS in 9m 51s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/4/check/openstack-tox-py27/c67b6ac/ : SUCCESS in 12m 01s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/4/check/openstack-tox-py35/d0eceb1/ : SUCCESS in 13m 25s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/4/check/openstack-tox-py36/1e36ab1/ : SUCCESS in 16m 13s\n- openstack-tox-docs http://logs.openstack.org/82/639382/4/check/openstack-tox-docs/077cc96/html/ : SUCCESS in 7m 16s\n- nova-tox-functional http://logs.openstack.org/82/639382/4/check/nova-tox-functional/849fb34/ : SUCCESS in 17m 09s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/4/check/nova-tox-functional-py35/7cef47c/ : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d5d216e4173d509e7f1a909a0fbf2be7942e7dc2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-14 22:09:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/4/check/EMC_VxFlexOS_NOVA/0d9dbb9/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"ae565c4e659c617adbd1a439b5b552affa5ad832","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-14 22:42:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc89386ecb2f2d7acc9e1b1baf703f1d692afd75","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-14 23:42:05.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"573f33afceb5aab43305cd815c957ea59c39fdc6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-15 04:10:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/4 : SUCCESS in 2h 19m 15s","accounts_in_message":[],"_revision_number":4},{"id":"68daafa8bd6d123d527af69c956a3b6b6fd5aedf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-15 11:11:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ecb148e69181bda8479b5d47f25a2afb42b8c562","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-15 11:12:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9a1ed83e79c14359a4dca605a8342fe4f75941fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-15 12:10:30.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/5/check-vote/ext-nova-zuul/91b8468 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":5},{"id":"fa86005ae2f1101832b6cd7dd7bb007083943d13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 12:36:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/5/check/openstack-tox-cover/e7a2b82/cover/ : SUCCESS in 15m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/5/check/openstack-tox-lower-constraints/879863d/ : SUCCESS in 12m 45s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/5/check/openstack-tox-pep8/ca1dfa6/ : SUCCESS in 11m 14s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/5/check/openstack-tox-py27/5f840df/ : SUCCESS in 15m 06s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/5/check/openstack-tox-py35/ea07420/ : SUCCESS in 14m 02s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/5/check/openstack-tox-py36/65081f1/ : SUCCESS in 14m 10s\n- openstack-tox-docs http://logs.openstack.org/82/639382/5/check/openstack-tox-docs/47928b7/html/ : SUCCESS in 7m 38s\n- nova-tox-functional http://logs.openstack.org/82/639382/5/check/nova-tox-functional/8262657/ : SUCCESS in 22m 18s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/5/check/nova-tox-functional-py35/96bf2c6/ : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":5},{"id":"6a92c213c4ab8573cd2dd909c28d935cd06e6f3b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-15 13:35:16.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29210 : FAILURE in 1h 08m 47s","accounts_in_message":[],"_revision_number":5},{"id":"4b968de387b0a84911e29cd64ef4c4cde721f2d4","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-15 13:38:08.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/5 : FAILURE in 2h 24m 08s","accounts_in_message":[],"_revision_number":5},{"id":"4f8b99034ec98644f67d20a0b30a723853255d51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-16 06:29:46.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"a2b08a92c922707da60cede466be082c091bb37f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-26 19:54:09.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nYour patches always seem to do \"a few things\".\n\nMost of the time it makes sense, but here I think the new helper method ought to be moved out to where it\u0027s used so it\u0027s not dead code here.","accounts_in_message":[],"_revision_number":5},{"id":"d193862fbf4663235b218d9bb1b7774d18e0192e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-01 13:01:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3a8fc9a77ff91a80a4df1e8422a42486fa57b73","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-01 16:54:58.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"5d6a16863f114fe75745943f32b589ffe987ce89","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-01 22:16:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"41f5b3c1f5193040ba525632e02697e840de54bf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-01 22:18:08.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2b83c3d70ee998e4386fa8c3fe010a4890db9356","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 22:18:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aaa62a3cfbc97e57256feddb6f4d55b58b0244ad","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-01 22:32:15.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ab8cf8a8956fd505435cda79e8fe9b0fa47311c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 23:01:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/6/check/openstack-tox-cover/dbc330a/cover/ : SUCCESS in 14m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/6/check/openstack-tox-lower-constraints/fa99699/ : SUCCESS in 17m 17s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/6/check/openstack-tox-pep8/9ec3cb2/ : SUCCESS in 12m 33s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/6/check/openstack-tox-py27/f7f6c87/ : SUCCESS in 15m 10s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/6/check/openstack-tox-py35/33df2f7/ : SUCCESS in 26m 23s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/6/check/openstack-tox-py36/31dd5bb/ : SUCCESS in 12m 45s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/6/check/openstack-tox-py37/93a7b13/ : SUCCESS in 13m 39s\n- openstack-tox-docs http://logs.openstack.org/82/639382/6/check/openstack-tox-docs/2a9651d/html/ : SUCCESS in 10m 13s\n- nova-tox-functional http://logs.openstack.org/82/639382/6/check/nova-tox-functional/5085422/ : SUCCESS in 25m 06s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/6/check/nova-tox-functional-py35/11971dc/ : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":6},{"id":"566189f6a9ff790785542c8dc10cb25e4fae7182","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 23:01:41.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/6/check/EMC_VxFlexOS_NOVA/547e1ff/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"6255b34e3571cf63d350d3ec1e811aba917d732d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-02 00:57:47.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"878e5c5a443a1dfb514df0c1befcbfbf5befd730","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 02:36:29.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29806 : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":6},{"id":"9020c1336972404b3753867a286e388c8e6c9c35","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 04:55:24.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/6 : FAILURE in 2h 20m 05s","accounts_in_message":[],"_revision_number":6},{"id":"bb196dd9f86037a0ba4180d2342464e8b102b167","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-03 01:42:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3d5d9f1eddf73821f22889987698cd766a2acf57","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-03 08:34:07.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/6/check-vote/ext-nova-zuul/292a1ce : FAILURE in 1h 22m 04s","accounts_in_message":[],"_revision_number":6},{"id":"aebc8b10c6c40b1ec5b7f0ce2826629cde29207d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-03 19:08:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5de2ed31b4937a04f633f3b04c058f0f2251c209","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-03 21:11:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b0ec729cf67518ced8c6b8162bd87f18b18f2150","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-03 21:14:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aaf50e6cfef0f4d31080ed825192903c991ff2a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-03 21:14:39.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"3cac2d267672fd9806cd0b4511fb464173db8732","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-03 21:35:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d714480c892123997a0d38ac29aaf272f1265d1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-03 22:48:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/7/check/openstack-tox-cover/095d2a5/cover/ : SUCCESS in 16m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/7/check/openstack-tox-lower-constraints/78afe0d/ : SUCCESS in 14m 16s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/7/check/openstack-tox-pep8/5bee239/ : SUCCESS in 11m 04s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/7/check/openstack-tox-py27/f889fe0/ : SUCCESS in 15m 10s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/7/check/openstack-tox-py35/40a3baf/ : SUCCESS in 14m 07s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/7/check/openstack-tox-py36/7d56162/ : SUCCESS in 13m 31s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/7/check/openstack-tox-py37/90930fb/ : SUCCESS in 16m 02s\n- openstack-tox-docs http://logs.openstack.org/82/639382/7/check/openstack-tox-docs/fb7267e/html/ : SUCCESS in 9m 35s\n- nova-tox-functional http://logs.openstack.org/82/639382/7/check/nova-tox-functional/6e5e275/ : SUCCESS in 18m 33s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/7/check/nova-tox-functional-py35/35b7f56/ : SUCCESS in 24m 18s","accounts_in_message":[],"_revision_number":7},{"id":"22902f48c55ba05c0f51f17a732eda111816beb4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-03 22:49:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/7/check/EMC_VxFlexOS_NOVA/df918a8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"1b6bffdb1b963ff9a5ce9088446ddc2139a51469","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-04 00:04:45.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/7 : FAILURE in 2h 49m 52s","accounts_in_message":[],"_revision_number":7},{"id":"dc8de55a1e2ad6682dc5dac358064a1baee55a25","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-04 00:13:50.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"2c8c64f7fd88b3a69076d7e2b4771ff007cc0e28","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-04 01:36:21.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29962 : SUCCESS in 1h 29m 47s","accounts_in_message":[],"_revision_number":7},{"id":"c224fe431ce0fce44b26c9935432cfeb5108ac55","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-06 01:04:33.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"794ed0af7c28427ba14af06ddaa91d5530fb350d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-06 01:41:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/8/check/openstack-tox-cover/2087391/cover/ : SUCCESS in 25m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/8/check/openstack-tox-lower-constraints/0cc597c/ : SUCCESS in 24m 22s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/8/check/openstack-tox-pep8/44d6a59/ : SUCCESS in 10m 17s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/8/check/openstack-tox-py27/1c81915/ : SUCCESS in 15m 35s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/8/check/openstack-tox-py35/93a5542/ : SUCCESS in 15m 51s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/8/check/openstack-tox-py36/97415ea/ : SUCCESS in 14m 56s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/8/check/openstack-tox-py37/329d7c8/ : SUCCESS in 23m 31s\n- openstack-tox-docs http://logs.openstack.org/82/639382/8/check/openstack-tox-docs/e28c7ae/html/ : SUCCESS in 7m 28s\n- nova-tox-functional http://logs.openstack.org/82/639382/8/check/nova-tox-functional/a3f739b/ : SUCCESS in 23m 59s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/8/check/nova-tox-functional-py35/83e0c8a/ : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":8},{"id":"d95aa7fa0dbea4094e2d04e90d2d00269834f0df","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-06 01:42:30.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/8/check/EMC_VxFlexOS_NOVA/6a83a3e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"26c130b84905921b1024a123c1b3ea2082f9e150","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-06 02:53:18.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"61af32ebe214b9a0327ddefa17db5820f0e99a7d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-06 03:24:50.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/8 : SUCCESS in 2h 17m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7fba80ddf497434a648d38bbecd919c83a92da3b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-06 07:04:49.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30128 : SUCCESS in 2h 48m 47s","accounts_in_message":[],"_revision_number":8},{"id":"48b4de8438380b8035a01f1758c4d15ce427114d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-06 10:32:10.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/8/check-vote/ext-nova-zuul/081ad4a : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":8},{"id":"c408721cc3a24c9af87ac403afd2e250a2fb7fe3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-11 19:21:35.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"f627344462094ef865fe5bdf121ff2393d17342b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 19:58:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/9/check/openstack-tox-cover/bace3c7/cover/ : SUCCESS in 16m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/9/check/openstack-tox-lower-constraints/3e0e9b4/ : SUCCESS in 14m 39s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/9/check/openstack-tox-pep8/2213044/ : SUCCESS in 10m 28s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/9/check/openstack-tox-py27/8af2ad5/ : SUCCESS in 13m 35s\n- openstack-tox-py35 http://logs.openstack.org/82/639382/9/check/openstack-tox-py35/a58ea55/ : SUCCESS in 14m 04s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/9/check/openstack-tox-py36/91266b6/ : SUCCESS in 12m 20s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/9/check/openstack-tox-py37/6df1d33/ : SUCCESS in 13m 03s\n- openstack-tox-docs http://logs.openstack.org/82/639382/9/check/openstack-tox-docs/d2eac53/html/ : SUCCESS in 9m 10s\n- nova-tox-functional http://logs.openstack.org/82/639382/9/check/nova-tox-functional/beb2418/ : SUCCESS in 20m 16s\n- nova-tox-functional-py35 http://logs.openstack.org/82/639382/9/check/nova-tox-functional-py35/9e60f68/ : SUCCESS in 22m 30s","accounts_in_message":[],"_revision_number":9},{"id":"2a5df895e4d7c1084effa9d50d1b0802c235d41a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 19:59:58.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/9/check/EMC_VxFlexOS_NOVA/c92eb71/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"1d6e6a1041a8e1fc326173c36ae53bb6a2f1fde9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-11 20:21:15.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/82/639382/9/check-vote/ext-nova-zuul/885eb0e : FAILURE in 57m 28s","accounts_in_message":[],"_revision_number":9},{"id":"3ba90fa93832301f33901f114eac97fa59c025cb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-11 21:06:19.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"08f3961e5e3d6783e1dedc7979327b1870b5b49e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-11 21:42:18.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/9 : FAILURE in 2h 17m 40s","accounts_in_message":[],"_revision_number":9},{"id":"65914bab838427efdb5dcb6f95921482728d3e52","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-11 23:48:57.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30351 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":9},{"id":"f78dacf8e0f375d956e5a75faa12b11565494b03","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-17 22:01:39.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"25e9e843bfb0a4527345844687eae6e35900930b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 22:03:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dc9333b16c159075d8a939ccb29dc332a1b863ed","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-18 00:35:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/10 : SUCCESS in 2h 30m 27s","accounts_in_message":[],"_revision_number":10},{"id":"1868b203d8390595dd833b37f52b5a47e12906df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 04:50:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/10/check/openstack-tox-cover/34c10b9/cover/ : SUCCESS in 15m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/10/check/openstack-tox-lower-constraints/cd577ae/ : SUCCESS in 15m 37s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/10/check/openstack-tox-pep8/3efa51d/ : SUCCESS in 11m 06s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/10/check/openstack-tox-py27/f51c39d/ : SUCCESS in 14m 02s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/10/check/openstack-tox-py36/2184168/ : SUCCESS in 16m 43s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/10/check/openstack-tox-py37/82a3624/ : SUCCESS in 14m 06s\n- openstack-tox-docs http://logs.openstack.org/82/639382/10/check/openstack-tox-docs/51d981c/html/ : SUCCESS in 6m 37s\n- nova-tox-functional http://logs.openstack.org/82/639382/10/check/nova-tox-functional/56beaf8/ : SUCCESS in 21m 15s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639382/10/check/nova-tox-functional-py36/e3344ba/ : SUCCESS in 20m 16s","accounts_in_message":[],"_revision_number":10},{"id":"b33bc4685623ecf5506c98e98adc9cd5e58fc47f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-18 04:51:15.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/10/check/EMC_VxFlexOS_NOVA/95115b5/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"626f20c3e964f3d4722f9fdc645494c9411f6875","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-18 05:45:14.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30650 : SUCCESS in 2h 41m 57s","accounts_in_message":[],"_revision_number":10},{"id":"9e2ccd9081c15fd298b093d9e2f47cb9f42509aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-18 05:54:59.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"c7b3c5f78d0e179f322d568897c5d8cb3f258fb4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-19 01:00:13.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"81e36b959c3df41cde5a8eb1f8cb523ec7d7745c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-19 01:02:16.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"09a09eafb7354d26c6584291fcaa49d5a47226fd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-19 03:19:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639382/11 : SUCCESS in 2h 15m 33s","accounts_in_message":[],"_revision_number":11},{"id":"cec9200f8dc7df4923e827ad9138fb442a64fdb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 05:59:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/11/check/openstack-tox-cover/87f82b9/cover/ : SUCCESS in 17m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/11/check/openstack-tox-lower-constraints/37f1eac/ : SUCCESS in 17m 33s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/11/check/openstack-tox-pep8/33d7fa9/ : SUCCESS in 11m 20s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/11/check/openstack-tox-py27/8caf433/ : SUCCESS in 13m 11s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/11/check/openstack-tox-py36/f670bca/ : SUCCESS in 13m 02s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/11/check/openstack-tox-py37/5f9ceed/ : SUCCESS in 14m 20s\n- openstack-tox-docs http://logs.openstack.org/82/639382/11/check/openstack-tox-docs/6f49ae4/html/ : SUCCESS in 7m 42s\n- nova-tox-functional http://logs.openstack.org/82/639382/11/check/nova-tox-functional/76ffae3/ : SUCCESS in 21m 01s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639382/11/check/nova-tox-functional-py36/a80bf20/ : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":11},{"id":"96081c8bccf155e39598af73d4500ad4a95b25c7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-19 05:59:27.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/11/check/EMC_VxFlexOS_NOVA/fc4d102/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"1f264472ac358befad88510f47be6e2f26bc018e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-19 07:05:43.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-11 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":11},{"id":"867db59ca82e227cc9af5dfdab9ab6097bdb0194","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-19 07:18:27.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30734 : FAILURE in 1h 19m 26s","accounts_in_message":[],"_revision_number":11},{"id":"75b295b2584d30ccb86acf023ba082d44394caf2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-19 15:03:22.000000000","message":"Patch Set 11:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639382/11/check/nova-out-of-tree-pvm/3f7e39b : FAILURE in 15m 00s\n- nova-in-tree-pvm http://184.172.12.213/82/639382/11/check/nova-in-tree-pvm/9fecb28 : FAILURE in 2h 24m 33s","accounts_in_message":[],"_revision_number":11},{"id":"a49cfdd6e49769f217f4f75986df3bf9a345ac09","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-20 22:13:39.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"a588660a68656c695595580c35ae4a6724706b82","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-20 22:15:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bd79abe8329b6f0f81d80bfdc58b1f5bca9844c9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-20 22:28:13.000000000","message":"Patch Set 12:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/82/639382/12/check/nova-out-of-tree-pvm/12500be : FAILURE in 12m 06s\n- nova-in-tree-pvm http://184.172.12.213/82/639382/12/check/nova-in-tree-pvm/2638879 : FAILURE in 1m 37s","accounts_in_message":[],"_revision_number":12},{"id":"e51578d381a8be6983b512859bcb3f4c0c9b15a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:44:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/82/639382/12/check/openstack-tox-cover/9449a79/cover/ : SUCCESS in 15m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/12/check/openstack-tox-lower-constraints/eec09f2/ : SUCCESS in 15m 06s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/12/check/openstack-tox-pep8/99b6f4b/ : SUCCESS in 11m 30s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/12/check/openstack-tox-py27/07b06c7/ : SUCCESS in 12m 32s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/12/check/openstack-tox-py36/345ab3c/ : SUCCESS in 11m 29s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/12/check/openstack-tox-py37/a42d7e1/ : SUCCESS in 12m 45s\n- openstack-tox-docs http://logs.openstack.org/82/639382/12/check/openstack-tox-docs/d678881/html/ : SUCCESS in 7m 18s\n- nova-tox-functional http://logs.openstack.org/82/639382/12/check/nova-tox-functional/ebc46c6/ : SUCCESS in 23m 24s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639382/12/check/nova-tox-functional-py36/0167e19/ : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":12},{"id":"dd4ae4a02d2a9c8ddfeb9ff1048bf51f74c666b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-20 22:44:58.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/82/639382/12/check/EMC_VxFlexOS_NOVA/a72ea11/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":12},{"id":"d620457a2a6aa371237a29c18e1e1944f02dfe3c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-20 22:45:17.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639382/12 : FAILURE in 20m 14s","accounts_in_message":[],"_revision_number":12},{"id":"6ce78175c77abdcd067d292f9f3d0e9377fe8891","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-20 23:09:33.000000000","message":"Patch Set 12:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-82-639382-12 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":12},{"id":"56688e479ff54d939dc7aa1663ed9aed7f5079e3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-21 03:59:35.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30797 : FAILURE in 3s","accounts_in_message":[],"_revision_number":12},{"id":"c89ddce182c4c5595ac70b7159601aaeb0bda6f1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-23 13:27:30.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"1e4d1223139b1cebc4d6a36bcd0ad8149dcbc88a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:08:36.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"9547f1719bcc0365c759ede58a2d19de54798ee3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:54:41.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/82/639382/12/gate/openstack-tox-lower-constraints/fd28773/ : SUCCESS in 16m 30s\n- openstack-tox-pep8 http://logs.openstack.org/82/639382/12/gate/openstack-tox-pep8/e2e0ce4/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/82/639382/12/gate/openstack-tox-py27/0a1129a/ : SUCCESS in 11m 47s\n- openstack-tox-py36 http://logs.openstack.org/82/639382/12/gate/openstack-tox-py36/e54d512/ : SUCCESS in 17m 00s\n- openstack-tox-py37 http://logs.openstack.org/82/639382/12/gate/openstack-tox-py37/a226859/ : SUCCESS in 11m 34s\n- openstack-tox-docs http://logs.openstack.org/82/639382/12/gate/openstack-tox-docs/2a972b2/html/ : SUCCESS in 8m 39s\n- nova-tox-functional http://logs.openstack.org/82/639382/12/gate/nova-tox-functional/a2f042e/ : SUCCESS in 18m 46s\n- nova-tox-functional-py36 http://logs.openstack.org/82/639382/12/gate/nova-tox-functional-py36/13dd443/ : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":12},{"id":"6ff283c600f31260375fc6ca805db7bce08b9eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:54:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"576a4e5260c4bc813474dead4eec19bd2a1cc680","revisions":{"eee94b87b096eeefed9a15e550690af7c9cfaa59":{"kind":"REWORK","_number":1,"created":"2019-02-26 16:40:42.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/1"}}},"commit":{"parents":[{"commit":"162434cbe58cb01ee5ee4ffe7b954019af0e03f8","subject":"Add cross-cell resize policy rule and enable in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/162434cbe58cb01ee5ee4ffe7b954019af0e03f8"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does a few things:\n\n1. Add a attachment_ids_for_instance() method which\n   can be used to get all volume attachment IDs for an\n   instance which is useful in tests that need to make\n   sure attachments are tracked and cleaned up properly,\n   i.e. to make sure nothing leaked.\n\n2. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n3. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eee94b87b096eeefed9a15e550690af7c9cfaa59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eee94b87b096eeefed9a15e550690af7c9cfaa59"}]},"branch":"refs/heads/master"},"8719fb2f8752f3480c7da79a53e3d5b013155e4d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-02-26 22:16:41.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/2"}}},"commit":{"parents":[{"commit":"debff531a8f0906cbdd38fbbb75e4c351d43412a","subject":"Stub out port binding create/delete in NeutronFixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/debff531a8f0906cbdd38fbbb75e4c351d43412a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-26 17:48:50.000000000","tz":-300},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does a few things:\n\n1. Add a attachment_ids_for_instance() method which\n   can be used to get all volume attachment IDs for an\n   instance which is useful in tests that need to make\n   sure attachments are tracked and cleaned up properly,\n   i.e. to make sure nothing leaked.\n\n2. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n3. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8719fb2f8752f3480c7da79a53e3d5b013155e4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8719fb2f8752f3480c7da79a53e3d5b013155e4d"}]},"branch":"refs/heads/master"},"1e85fdf085f36b7c2b1313bd08decbdeaf3d6858":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-07 22:59:36.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/3"}}},"commit":{"parents":[{"commit":"48324a68149fda07c0c07445fc8f50b9d5e665db","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48324a68149fda07c0c07445fc8f50b9d5e665db"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-03-07 22:35:52.000000000","tz":-300},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does a few things:\n\n1. Add a attachment_ids_for_instance() method which\n   can be used to get all volume attachment IDs for an\n   instance which is useful in tests that need to make\n   sure attachments are tracked and cleaned up properly,\n   i.e. to make sure nothing leaked.\n\n2. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n3. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e85fdf085f36b7c2b1313bd08decbdeaf3d6858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e85fdf085f36b7c2b1313bd08decbdeaf3d6858"}]},"branch":"refs/heads/master"},"594828dcdbea36607497766d0c121a58173729cc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-14 20:34:04.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/4"}}},"commit":{"parents":[{"commit":"f10c5bb11d7f47d6bc0d5f24e252c3c25640d913","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f10c5bb11d7f47d6bc0d5f24e252c3c25640d913"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-03-14 20:32:54.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does a few things:\n\n1. Add a attachment_ids_for_instance() method which\n   can be used to get all volume attachment IDs for an\n   instance which is useful in tests that need to make\n   sure attachments are tracked and cleaned up properly,\n   i.e. to make sure nothing leaked.\n\n2. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n3. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/594828dcdbea36607497766d0c121a58173729cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/594828dcdbea36607497766d0c121a58173729cc"}]},"branch":"refs/heads/master"},"5418e8ea6688ca4c717d8b511720a20047656a1f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-15 11:11:12.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/5"}}},"commit":{"parents":[{"commit":"51aa230d2f0afe8a8fa9c5fb58f187ef1b243371","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51aa230d2f0afe8a8fa9c5fb58f187ef1b243371"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-03-15 11:10:43.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does a few things:\n\n1. Add a attachment_ids_for_instance() method which\n   can be used to get all volume attachment IDs for an\n   instance which is useful in tests that need to make\n   sure attachments are tracked and cleaned up properly,\n   i.e. to make sure nothing leaked.\n\n2. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n3. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5418e8ea6688ca4c717d8b511720a20047656a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5418e8ea6688ca4c717d8b511720a20047656a1f"}]},"branch":"refs/heads/master"},"bd2e53d03997e77f5cdc2f930ddacef0e0f48d95":{"kind":"REWORK","_number":6,"created":"2019-04-01 22:16:52.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/6"}}},"commit":{"parents":[{"commit":"239cd895bc45ab7a214d80c6f184b6c7a2cdd45a","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/239cd895bc45ab7a214d80c6f184b6c7a2cdd45a"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-01 22:04:48.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does two things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd2e53d03997e77f5cdc2f930ddacef0e0f48d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd2e53d03997e77f5cdc2f930ddacef0e0f48d95"}]},"branch":"refs/heads/master"},"c58af65bbbd9252f5eefb4aa808f842173825fa2":{"kind":"REWORK","_number":7,"created":"2019-04-03 21:11:57.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/7"}}},"commit":{"parents":[{"commit":"a4a686470bd18a73803b0537e1209eab7e236f01","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4a686470bd18a73803b0537e1209eab7e236f01"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-03 18:50:59.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c58af65bbbd9252f5eefb4aa808f842173825fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c58af65bbbd9252f5eefb4aa808f842173825fa2"}]},"branch":"refs/heads/master"},"f8c08322ed268e745f5385cfb3d156611154c338":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-06 01:04:33.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/8"}}},"commit":{"parents":[{"commit":"1a997b39b052d2b711728c4eeb2e4bdd884984c9","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a997b39b052d2b711728c4eeb2e4bdd884984c9"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-06 00:50:26.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8c08322ed268e745f5385cfb3d156611154c338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8c08322ed268e745f5385cfb3d156611154c338"}]},"branch":"refs/heads/master"},"9b4c1d518128a32b319b0ac160fe7873ddfe8a26":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-04-11 19:21:35.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/9"}}},"commit":{"parents":[{"commit":"f45928b1584ec2b429adc88fd22b50fb47a9f792","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f45928b1584ec2b429adc88fd22b50fb47a9f792"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-11 18:00:32.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b4c1d518128a32b319b0ac160fe7873ddfe8a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b4c1d518128a32b319b0ac160fe7873ddfe8a26"}]},"branch":"refs/heads/master"},"531439b03ef988c0e2a673fa7ff1d441ec469aef":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-04-17 22:01:39.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/10"}}},"commit":{"parents":[{"commit":"c4d78874d286acb2c88fc13064e70de2ac53d5cf","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4d78874d286acb2c88fc13064e70de2ac53d5cf"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-17 19:22:17.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/531439b03ef988c0e2a673fa7ff1d441ec469aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/531439b03ef988c0e2a673fa7ff1d441ec469aef"}]},"branch":"refs/heads/master"},"1cacf0cdcbc11861364878bf8c3464056ec95b14":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-04-19 01:00:13.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/11"}}},"commit":{"parents":[{"commit":"89e9e70da7735471778a5bc037e1101833b9e0c9","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89e9e70da7735471778a5bc037e1101833b9e0c9"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-18 22:32:46.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cacf0cdcbc11861364878bf8c3464056ec95b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cacf0cdcbc11861364878bf8c3464056ec95b14"}]},"branch":"refs/heads/master"},"576a4e5260c4bc813474dead4eec19bd2a1cc680":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-20 22:13:39.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/82/639382/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/639382/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/639382/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/639382/12"}}},"commit":{"parents":[{"commit":"4df3f17853c2be1bf4533dfff0f7298e2cf1b8d0","subject":"Fix ProviderUsageBaseTestCase._run_periodics for multi-cell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4df3f17853c2be1bf4533dfff0f7298e2cf1b8d0"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-25 23:52:41.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-20 22:13:07.000000000","tz":-240},"subject":"Improve CinderFixtureNewAttachFlow","message":"Improve CinderFixtureNewAttachFlow\n\nThis does three things:\n\n1. Implements the attachment_complete method in order to\n   fail if code is trying to complete an attachment that\n   does not exist.\n\n2. Adds logging in the various attachment CRUD methods to\n   aid in debugging test failures.\n\n3. Mirrors the method signature for is_microversion_supported\n   to make sure code is at least calling it properly.\n\nChange-Id: If6a36d23768877bfa820ed44b610bfb113df5464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/576a4e5260c4bc813474dead4eec19bd2a1cc680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/576a4e5260c4bc813474dead4eec19bd2a1cc680"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
