)]}'
{"id":"openstack%2Fnova~639396","triplet_id":"openstack%2Fnova~master~I125e2af11ebcdc444c62ed41a9ce7985db6d5576","project":"openstack/nova","branch":"master","topic":"bug/1813789","hashtags":[],"change_id":"I125e2af11ebcdc444c62ed41a9ce7985db6d5576","subject":"Revert \"Wait for network-vif-plugged on resize revert\"","status":"ABANDONED","created":"2019-02-26 17:37:50.000000000","updated":"2019-06-06 06:53:23.000000000","total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4d1ade3d2bf098d8d07232deaa337b1c0b40a1e","_number":639396,"virtual_id_number":639396,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-04-29 16:32:35.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"date":"2019-04-29 14:04:44.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"date":"2019-04-29 14:05:59.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2019-04-29 16:55:10.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-29 15:17:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2019-04-29 14:19:47.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-04-29 15:54:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":1,"date":"2019-04-29 16:54:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-04-29 18:04:50.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2019-04-29 14:53:48.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2019-04-29 16:16:04.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"date":"2019-04-29 13:59:25.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"date":"2019-06-06 06:53:23.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-08 15:07:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-03-21 12:19:47.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-03-26 20:09:39.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-17 13:21:24.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-27 15:07:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 13:59:25.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 14:04:44.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 14:05:59.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 14:19:47.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 14:53:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 15:17:16.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 15:54:01.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-04-29 16:16:04.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-04-29 16:32:35.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 16:54:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 16:55:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 18:04:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-06 06:53:23.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"35f4171a00d1bda84dcf4228df07c10373916117","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-02-26 17:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6c0d0c3e3437396f1dd5d1ae7c348db72ebe7c2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 17:38:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e39e1f2848e1221bfd27d08dc60110a057928a57","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-02-26 17:45:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/1 : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"50d5b627730e62ec686226c55d42a9a61c163adb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 19:01:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/1/check/grenade-py3/5247c06/ : FAILURE in 48m 36s\n- tempest-full-py3 http://logs.openstack.org/96/639396/1/check/tempest-full-py3/09ee409/ : FAILURE in 28m 40s\n- openstack-tox-cover http://logs.openstack.org/96/639396/1/check/openstack-tox-cover/6dbe005/ : FAILURE in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/1/check/openstack-tox-lower-constraints/d6f8a4f/ : FAILURE in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/1/check/openstack-tox-pep8/253b328/ : FAILURE in 11m 08s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/1/check/openstack-tox-py27/904c839/ : FAILURE in 6m 21s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/1/check/openstack-tox-py35/385b1e9/ : FAILURE in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/1/check/openstack-tox-py36/d07d9a7/ : FAILURE in 6m 15s\n- openstack-tox-docs http://logs.openstack.org/96/639396/1/check/openstack-tox-docs/f23da12/html/ : SUCCESS in 6m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c54daf4/ : FAILURE in 41m 47s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/1/check/devstack-plugin-ceph-tempest/4e78fb5/ : FAILURE in 34m 32s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/96/639396/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/361e3e1/ : FAILURE in 1h 07m 52s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/1/check/neutron-grenade-multinode/6099039/ : FAILURE in 1h 05m 55s\n- nova-live-migration http://logs.openstack.org/96/639396/1/check/nova-live-migration/f36542e/ : FAILURE in 29m 37s\n- nova-next http://logs.openstack.org/96/639396/1/check/nova-next/6f65dff/ : FAILURE in 36m 46s\n- nova-tox-functional http://logs.openstack.org/96/639396/1/check/nova-tox-functional/8cb8060/ : FAILURE in 5m 47s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/1/check/nova-tox-functional-py35/e48c691/ : FAILURE in 6m 01s\n- tempest-multinode-full http://logs.openstack.org/96/639396/1/check/tempest-multinode-full/ac2607a/ : FAILURE in 27m 17s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/96/639396/1/check/tempest-slow-py3/9477ecd/ : FAILURE in 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"053f68a35a0433e6598c7b2837865300039d4d0b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-26 19:08:07.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28291 : FAILURE in 1h 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"025e9c444f88db637804493247af379fa3addde6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-27 00:41:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/1/check-vote/ext-nova-zuul/f7da7e0 : FAILURE in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4038e7c7696c6188cbceafd80d2a50c81c713800","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-02-27 14:57:13.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/1/check/tempest-dsvm-full-xenial/c6cb4c1/ : FAILURE in 21m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/1/check/tempest-dsvm-full-xenial-py3/e0bf33a/ : FAILURE in 16m 44s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/1/check/grenade-dsvm-xenial/2d32e36/ : FAILURE in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f237f4b6861427c9caa9bbdcfae1fd75b8aca0f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 14:59:41.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/0676267 : FAILURE in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f04bd8ec5fabac310417e614d6c6392c70593608","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-02-28 16:21:11.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0d9e311992a032576f8d38ce2c1ef56fedd7c400","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-02-28 16:37:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nIt looks like maybe something has changed about what we\u0027re doing with neutron that ends up emitting the event. We should figure that out.\n\nHowever, I\u0027m massively against changing the way that this event system works, as it was designed to prevent the bug that this patch claims to be fixing, as well as some of the things that this patch introduces (i.e. not ever cleaning up for some events).\n\nThe goal of the context manager was to make it super clear where the critical section of code is that was doing the work that would end up with an event coming, and to avoid having to do an initial setup, with a bunch of cleanup exit paths. If we are poking code earlier that causes neutron to emit the notification, then I think we just need to move the context manager a little higher in the stack, not split it into two pieces.","accounts_in_message":[],"_revision_number":1},{"id":"5ef7ddf135849b13d7661a6a19916e6539666f33","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-02-28 17:03:54.000000000","message":"Patch Set 1:\n\n\u003e (3 comments)\n \u003e \n \u003e It looks like maybe something has changed about what we\u0027re doing\n \u003e with neutron that ends up emitting the event. We should figure that\n \u003e out.\n\nI think over time we\u0027ve just added so much stuff that by the time the code that actually listens for events is reached, Neutron has done its thing and has sent us the event. The downstream bug [1] maybe makes it clearer.\n\nYou\u0027re right, introducing state and cleaning up after is a pain, but I don\u0027t see a way around the fundamental necessity of listening as soon as we poke Neutron in way that\u0027ll cause it to send the event.\n\nMaybe then just don\u0027t change wait_for_instance_events() and just wrap it around self.network_api.setup_networks_on_host? Do we have all we need at that time to call it properly?\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1678681","accounts_in_message":[],"_revision_number":1},{"id":"362c08ca45e4ddb0ace2a074c3f2d8768a31d18d","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-02-28 19:03:33.000000000","message":"Patch Set 1:\n\n\u003e \u003e (3 comments)\n \u003e \u003e\n \u003e \u003e It looks like maybe something has changed about what we\u0027re doing\n \u003e \u003e with neutron that ends up emitting the event. We should figure\n \u003e that\n \u003e \u003e out.\n \u003e \n \u003e I think over time we\u0027ve just added so much stuff that by the time\n \u003e the code that actually listens for events is reached, Neutron has\n \u003e done its thing and has sent us the event. The downstream bug [1]\n \u003e maybe makes it clearer.\n \u003e \n \u003e You\u0027re right, introducing state and cleaning up after is a pain,\n \u003e but I don\u0027t see a way around the fundamental necessity of listening\n \u003e as soon as we poke Neutron in way that\u0027ll cause it to send the\n \u003e event.\n \u003e \n \u003e Maybe then just don\u0027t change wait_for_instance_events() and just\n \u003e wrap it around self.network_api.setup_networks_on_host? Do we have\n \u003e all we need at that time to call it properly?\n \u003e \n \u003e [1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d1678681\n\nOr I guess if we\u0027re going to introduce state, do it \"properly\" and straight up save external events in the database. Cleanup then becomes a matter of a periodic job that deletes older than X. It\u0027s a smidge less ugly than my current proposal, at the expense of extra DB queries.","accounts_in_message":[],"_revision_number":1},{"id":"0b7b5dfec5248ac78c1ea6e3544961082b09453e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-07 20:11:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63ff536466987942ab6bcf8a7b6ce42653c44d84","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-07 21:18:21.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28911 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":2},{"id":"12fa4c9920aef3d79e18137e5240b250f7d0b122","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 22:41:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/2 : FAILURE in 2h 29m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3891d3c6079cb71e44b9dac45489b6cbef643023","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 23:32:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/2/check/grenade-py3/0ba9e9b/ : SUCCESS in 1h 18m 57s\n- tempest-full-py3 http://logs.openstack.org/96/639396/2/check/tempest-full-py3/545eb26/ : SUCCESS in 1h 28m 58s\n- openstack-tox-cover http://logs.openstack.org/96/639396/2/check/openstack-tox-cover/e563931/ : FAILURE in 16m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/2/check/openstack-tox-lower-constraints/0d500f8/ : FAILURE in 14m 21s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/2/check/openstack-tox-pep8/2050089/ : SUCCESS in 11m 54s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/2/check/openstack-tox-py27/715ee8c/ : FAILURE in 12m 43s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/2/check/openstack-tox-py35/9697aea/ : FAILURE in 12m 57s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/2/check/openstack-tox-py36/cc8b67a/ : FAILURE in 11m 46s\n- openstack-tox-docs http://logs.openstack.org/96/639396/2/check/openstack-tox-docs/7debc4d/html/ : SUCCESS in 6m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/56e9aa8/ : SUCCESS in 48m 41s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/2/check/devstack-plugin-ceph-tempest/df78b37/ : SUCCESS in 1h 29m 26s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/2/check/neutron-grenade-multinode/2a86898/ : SUCCESS in 1h 05m 17s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/2/check/nova-grenade-live-migration/7c3ad9d/ : SUCCESS in 58m 47s\n- nova-live-migration http://logs.openstack.org/96/639396/2/check/nova-live-migration/c0f23f6/ : SUCCESS in 51m 36s\n- nova-lvm http://logs.openstack.org/96/639396/2/check/nova-lvm/c1e4b15/ : SUCCESS in 1h 11m 25s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/2/check/nova-next/7a2a1ec/ : SUCCESS in 1h 59m 19s\n- nova-tox-functional http://logs.openstack.org/96/639396/2/check/nova-tox-functional/874c5be/ : FAILURE in 19m 51s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/2/check/nova-tox-functional-py35/ebbd00d/ : FAILURE in 20m 27s\n- tempest-multinode-full http://logs.openstack.org/96/639396/2/check/tempest-multinode-full/bdf39ec/ : SUCCESS in 1h 53m 23s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/96/639396/2/check/tempest-slow-py3/1d4b4de/ : SUCCESS in 2h 02m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b89197abb6be18827b83a33a73524cbce8fa6961","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-08 04:21:26.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/2/check-vote/ext-nova-zuul/43d6167 : FAILURE in 1h 02m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c9cd133e5c7288b5fc568c4359408758c5a7dacf","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 10:03:58.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c4a93bb : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4d0c723800dfcf8868bccd3fa73079bbc775b4ff","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-08 14:57:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d3f9eea6247166228d3eedbbf01a6022a282b0f3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-03-08 15:07:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"328df1b306adca243049869849895a12740bf9c0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 15:08:20.000000000","message":"Patch Set 3: Code-Review-1\n\nSo you\u0027re reverting https://review.openstack.org/#/c/595069/ but if you\u0027re going to do that you should definitely explain why that change was wrong given the details I put in the commit message there.","accounts_in_message":[],"_revision_number":3},{"id":"dfbbacd087dc3815ce32e3ef1e50c78e922b9f2d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 15:32:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ecd9c7a1745414637095b07e995e3c4fdd0a487","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 15:58:14.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28968 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e1de8e744ba3e4abb027faa7a649b6e946d3451a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-08 16:57:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"928b44096dab4a9c316bfec062d847087186daa9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-08 16:58:46.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"69b568a9b374c9a3b1729540f05372e6e0cec3c9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-08 16:58:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f88006849b23f58e8b6b2601171ee2dbeee7cab8","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-08 16:59:55.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1247719e4f32a20961bc791a913901f2cead4beb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 17:02:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a0ae9b0512350cf28a4b820934ff83eb085fb66","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 17:03:56.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11415d9321471d0291752fe7fbd6c91253f8df2e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 17:05:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a4c554464ee354570ce81fb15e831c422645f13a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-08 17:15:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"70d6a477c6d57b7d806e910fe0ee94a30ddadff5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 17:58:03.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28975 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":5},{"id":"a156ebebf962c22629dbb067c50efc7ee15eb368","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-08 18:47:19.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/5/check/tempest-dsvm-full-xenial/703d439/ : SUCCESS in 1h 46m 43s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/5/check/tempest-dsvm-full-xenial-py3/07ef5f2/ : SUCCESS in 1h 13m 31s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/5/check/grenade-dsvm-xenial/c38a4a0/ : SUCCESS in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"37489bf5cc527f8a02ec012e88033ddcc2315cf2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 21:40:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/5 : SUCCESS in 2h 09m 43s","accounts_in_message":[],"_revision_number":5},{"id":"bcf1f11d6cc60cba6638f320d4b6d506e3230bb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 21:50:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/5/check/grenade-py3/b256d52/ : SUCCESS in 55m 38s\n- tempest-full-py3 http://logs.openstack.org/96/639396/5/check/tempest-full-py3/a77443e/ : SUCCESS in 1h 29m 41s\n- openstack-tox-cover http://logs.openstack.org/96/639396/5/check/openstack-tox-cover/f9b09c6/cover/ : SUCCESS in 18m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/5/check/openstack-tox-lower-constraints/21085f7/ : SUCCESS in 12m 54s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/5/check/openstack-tox-pep8/114abda/ : SUCCESS in 10m 03s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/5/check/openstack-tox-py27/fd7750b/ : SUCCESS in 12m 43s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/5/check/openstack-tox-py35/fb61d3a/ : SUCCESS in 12m 33s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/5/check/openstack-tox-py36/a3bdde0/ : SUCCESS in 11m 42s\n- openstack-tox-docs http://logs.openstack.org/96/639396/5/check/openstack-tox-docs/489bfe7/html/ : SUCCESS in 6m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6e98f5b/ : SUCCESS in 49m 46s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/5/check/devstack-plugin-ceph-tempest/dc0a44d/ : FAILURE in 1h 30m 20s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/5/check/neutron-grenade-multinode/6a04a02/ : SUCCESS in 1h 09m 12s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/5/check/nova-grenade-live-migration/266d219/ : SUCCESS in 1h 06m 03s\n- nova-live-migration http://logs.openstack.org/96/639396/5/check/nova-live-migration/633094c/ : FAILURE in 48m 24s\n- nova-lvm http://logs.openstack.org/96/639396/5/check/nova-lvm/4a2d85a/ : SUCCESS in 1h 15m 03s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/5/check/nova-next/16c8df9/ : SUCCESS in 1h 59m 49s\n- nova-tox-functional http://logs.openstack.org/96/639396/5/check/nova-tox-functional/f71243a/ : SUCCESS in 19m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/5/check/nova-tox-functional-py35/9ff9408/ : SUCCESS in 19m 44s\n- tempest-multinode-full http://logs.openstack.org/96/639396/5/check/tempest-multinode-full/3a66f11/ : SUCCESS in 1h 48m 54s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/96/639396/5/check/tempest-slow-py3/71a457e/ : SUCCESS in 2h 21m 10s","accounts_in_message":[],"_revision_number":5},{"id":"aaaf3c12a08e6e7e5788124457ca4592bcdbeb5f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-09 10:44:19.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/5/check-vote/ext-nova-zuul/a9d04e4 : FAILURE in 1h 10m 19s","accounts_in_message":[],"_revision_number":5},{"id":"4d0efed80a5b712004fe42804d37b2e8fb56e048","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 13:39:36.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/91f62a8 : FAILURE in 24m 52s","accounts_in_message":[],"_revision_number":5},{"id":"74eddf456dd71d4e7af30288740ab9a2e5a71271","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-03-20 14:46:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5e9fb6d6cd78ee14773dfdffc0520d8923ba4118","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-20 14:47:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"028532f1162c3bfeca882d24081e33edfd47650d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-20 15:28:07.000000000","message":"Patch Set 6:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/6/check/nova-out-of-tree-pvm/5953d9a : FAILURE in 40m 31s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/6/check/nova-in-tree-pvm/0a1728d : FAILURE in 37m 09s","accounts_in_message":[],"_revision_number":6},{"id":"83f676f40c0f50b0af61a62d342c70501ff57433","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-20 16:11:57.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/6/check/tempest-dsvm-full-xenial/dd0b581/ : SUCCESS in 1h 12m 44s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/6/check/tempest-dsvm-full-xenial-py3/45c74e7/ : SUCCESS in 1h 24m 11s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/6/check/grenade-dsvm-xenial/89b2fc4/ : SUCCESS in 50m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"392d1d61ea6d494663f9271bf0dfdc8eb3aee5dc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-20 16:29:21.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29399 : SUCCESS in 1h 42m 01s","accounts_in_message":[],"_revision_number":6},{"id":"fbbf89cc43a49ef772257dc8e9bcc0bba20a453d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-20 16:33:08.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7c13660 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":6},{"id":"0f54c055adf686f416fe6de2859d20ec876fc5ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 17:16:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/6/check/grenade-py3/befefb3/ : SUCCESS in 51m 58s\n- tempest-full-py3 http://logs.openstack.org/96/639396/6/check/tempest-full-py3/b818b24/ : SUCCESS in 1h 26m 25s\n- openstack-tox-cover http://logs.openstack.org/96/639396/6/check/openstack-tox-cover/53575dc/cover/ : SUCCESS in 16m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/6/check/openstack-tox-lower-constraints/cad904e/ : SUCCESS in 13m 51s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/6/check/openstack-tox-pep8/af12772/ : SUCCESS in 9m 38s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/6/check/openstack-tox-py27/0ff1bfd/ : SUCCESS in 12m 51s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/6/check/openstack-tox-py35/cfb2657/ : SUCCESS in 13m 36s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/6/check/openstack-tox-py36/9b125a6/ : SUCCESS in 13m 04s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/6/check/openstack-tox-py37/872ad4c/ : SUCCESS in 13m 51s\n- openstack-tox-docs http://logs.openstack.org/96/639396/6/check/openstack-tox-docs/56bca30/html/ : SUCCESS in 7m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8f80a8a/ : SUCCESS in 56m 36s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/6/check/devstack-plugin-ceph-tempest/64c717e/ : SUCCESS in 1h 24m 51s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/6/check/neutron-grenade-multinode/5da5a00/ : SUCCESS in 1h 17m 18s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/6/check/nova-grenade-live-migration/cf38a81/ : SUCCESS in 1h 21m 09s\n- nova-live-migration http://logs.openstack.org/96/639396/6/check/nova-live-migration/f57043e/ : SUCCESS in 55m 26s\n- nova-lvm http://logs.openstack.org/96/639396/6/check/nova-lvm/90a3074/ : SUCCESS in 1h 21m 17s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/6/check/nova-next/017fd0c/ : SUCCESS in 1h 18m 07s\n- nova-tox-functional http://logs.openstack.org/96/639396/6/check/nova-tox-functional/7bb59b4/ : SUCCESS in 20m 06s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/6/check/nova-tox-functional-py35/9b657d3/ : SUCCESS in 17m 56s\n- tempest-multinode-full http://logs.openstack.org/96/639396/6/check/tempest-multinode-full/fa3ae83/ : SUCCESS in 1h 45m 55s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/96/639396/6/check/tempest-slow-py3/e1571c4/ : SUCCESS in 2h 22m 38s","accounts_in_message":[],"_revision_number":6},{"id":"6357238f7f3a2a3f84021831f92e7effa82b83ec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-20 17:17:21.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/6/check/EMC_VxFlexOS_NOVA/397648c/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"88763a9303b29720f9c0d39421f395a3e8c55c84","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-20 17:28:15.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/6 : FAILURE in 2h 40m 10s","accounts_in_message":[],"_revision_number":6},{"id":"100c95e427f617a1499c668d2b5e4671ef8727dd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-20 18:14:26.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"e2ed7dbe3d61da9c28a0daf98031c7d38b9fc23a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-26 20:09:39.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nBased on what Matt is saying, the subsequent patch should be squashed into this one (once it\u0027s ready) so we don\u0027t have a gap.","accounts_in_message":[],"_revision_number":6},{"id":"dad327b8226e7e4548ca9f8ce61808a60c6703d7","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-30 15:53:51.000000000","message":"Patch Set 6: Workflow-1\n\nI want to WIP this until the change on top is ready to go.","accounts_in_message":[],"_revision_number":6},{"id":"9f83801ad1a6958e9d3c6301b94f197415ec2a01","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-10 23:51:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c9a1903433543706b9314640f3aed229e0432ae9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-10 23:52:21.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"236e4713ba552a7eea9cb4447219da686d3ea2a7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-10 23:52:30.000000000","message":"Patch Set 7: Workflow-1\n\nRe-applying the -W after a rebase to resolve merge conflicts, as the patch on top is still WIP while I figure out why it\u0027s failing the test_revert_resize tempest tests.","accounts_in_message":[],"_revision_number":7},{"id":"a820da73afb8f7acf2893b1a4bed9b9cee04cf26","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-11 00:44:35.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/7/check-vote/ext-nova-zuul/05bf3af : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":7},{"id":"a91439a808723540cad3bb5c2bdf22903012f055","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-11 00:53:30.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30339 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":7},{"id":"6f3ca982f2c6f823f59379ecc37ea12b4f68d4c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-11 01:12:57.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/7/check/tempest-dsvm-full-xenial/45e5edf/ : SUCCESS in 1h 20m 22s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/7/check/tempest-dsvm-full-xenial-py3/ce0d51d/ : SUCCESS in 1h 16m 00s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/7/check/grenade-dsvm-xenial/72ab6b6/ : SUCCESS in 1h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b7b492073d5d6cfbbde7e771d6f9d55166a47982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 02:00:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/7/check/grenade-py3/edef789/ : SUCCESS in 52m 07s\n- tempest-full-py3 http://logs.openstack.org/96/639396/7/check/tempest-full-py3/6493197/ : SUCCESS in 1h 25m 58s\n- openstack-tox-cover http://logs.openstack.org/96/639396/7/check/openstack-tox-cover/081b6e2/cover/ : SUCCESS in 13m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/7/check/openstack-tox-lower-constraints/7d3f5be/ : SUCCESS in 13m 36s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/7/check/openstack-tox-pep8/676ff12/ : SUCCESS in 8m 59s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/7/check/openstack-tox-py27/446f7b8/ : SUCCESS in 11m 53s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/7/check/openstack-tox-py35/534bc6d/ : SUCCESS in 12m 34s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/7/check/openstack-tox-py36/86481fd/ : SUCCESS in 11m 24s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/7/check/openstack-tox-py37/ae499f0/ : SUCCESS in 12m 19s\n- openstack-tox-docs http://logs.openstack.org/96/639396/7/check/openstack-tox-docs/64a44ce/html/ : SUCCESS in 6m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a76019b/ : SUCCESS in 46m 34s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/7/check/devstack-plugin-ceph-tempest/a468c8b/ : SUCCESS in 1h 35m 41s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/7/check/neutron-grenade-multinode/e14f144/ : SUCCESS in 1h 01m 58s (non-voting)\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/7/check/nova-grenade-live-migration/27a7e03/ : SUCCESS in 59m 23s\n- nova-live-migration http://logs.openstack.org/96/639396/7/check/nova-live-migration/1b314f5/ : SUCCESS in 45m 01s\n- nova-lvm http://logs.openstack.org/96/639396/7/check/nova-lvm/b795a52/ : FAILURE in 1h 05m 06s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/7/check/nova-next/5075d8f/ : SUCCESS in 1h 19m 44s\n- nova-tox-functional http://logs.openstack.org/96/639396/7/check/nova-tox-functional/703edab/ : SUCCESS in 16m 55s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/7/check/nova-tox-functional-py35/83612d2/ : SUCCESS in 17m 25s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/7/check/tempest-slow-py3/23a9976/ : SUCCESS in 2h 04m 19s","accounts_in_message":[],"_revision_number":7},{"id":"cbee726c72fd4819d04ad08d50b6add9fec9f7f2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 02:01:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/7/check/EMC_VxFlexOS_NOVA/cd9d1ee/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"bae903df2d168da6b4d47333cd5a5f10b8d60b3c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-11 02:11:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/7 : SUCCESS in 2h 19m 25s","accounts_in_message":[],"_revision_number":7},{"id":"6c461fa9df3b38398e8b6b527947de3dbbae5d22","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-11 02:52:05.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"2b0323fb4ac32864681f9928477f8ecf343644bc","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-12 23:59:22.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ecf703864964ea45a4e504d40759d22225a5891b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-13 00:00:34.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"5c3adcbc45a74b92d119035a271d38995f4fc8e8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-13 01:12:52.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30462 : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":8},{"id":"fb883cc9ded8098c1d38ae62b14a2050bc1a6555","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-13 01:30:28.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/8/check/tempest-dsvm-full-xenial/863ff72/ : SUCCESS in 1h 20m 44s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/8/check/tempest-dsvm-full-xenial-py3/d083f9c/ : SUCCESS in 1h 29m 45s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/8/check/grenade-dsvm-xenial/5b418a1/ : SUCCESS in 1h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3c45503438ba90050876a643ef1511b23e496f5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 02:07:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/8/check/grenade-py3/eec61af/ : SUCCESS in 55m 46s\n- tempest-full-py3 http://logs.openstack.org/96/639396/8/check/tempest-full-py3/854876e/ : SUCCESS in 1h 33m 52s\n- openstack-tox-cover http://logs.openstack.org/96/639396/8/check/openstack-tox-cover/8c1f367/cover/ : SUCCESS in 17m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/8/check/openstack-tox-lower-constraints/9dc9a19/ : SUCCESS in 14m 39s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/8/check/openstack-tox-pep8/0d6713d/ : SUCCESS in 11m 31s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/8/check/openstack-tox-py27/906b20d/ : SUCCESS in 13m 36s\n- openstack-tox-py35 http://logs.openstack.org/96/639396/8/check/openstack-tox-py35/625527a/ : SUCCESS in 12m 08s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/8/check/openstack-tox-py36/e123d82/ : SUCCESS in 12m 30s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/8/check/openstack-tox-py37/65d29d4/ : SUCCESS in 15m 56s\n- openstack-tox-docs http://logs.openstack.org/96/639396/8/check/openstack-tox-docs/93de081/html/ : SUCCESS in 7m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f52fe27/ : SUCCESS in 49m 56s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/8/check/devstack-plugin-ceph-tempest/d22fabe/ : SUCCESS in 1h 23m 45s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/8/check/neutron-grenade-multinode/04edc20/ : SUCCESS in 1h 07m 47s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/8/check/nova-grenade-live-migration/aa318a2/ : SUCCESS in 1h 13m 37s\n- nova-live-migration http://logs.openstack.org/96/639396/8/check/nova-live-migration/7707dd0/ : SUCCESS in 51m 02s\n- nova-lvm http://logs.openstack.org/96/639396/8/check/nova-lvm/1940c38/ : SUCCESS in 1h 02m 36s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/8/check/nova-next/0810d5e/ : SUCCESS in 1h 10m 33s\n- nova-tox-functional http://logs.openstack.org/96/639396/8/check/nova-tox-functional/3aa70c3/ : SUCCESS in 20m 02s\n- nova-tox-functional-py35 http://logs.openstack.org/96/639396/8/check/nova-tox-functional-py35/91b51f3/ : SUCCESS in 17m 17s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/8/check/tempest-slow-py3/0d5fd95/ : SUCCESS in 1h 59m 33s","accounts_in_message":[],"_revision_number":8},{"id":"1292634409eae72dd2556b57f64ede21b31b8bda","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-13 02:08:37.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/8/check/EMC_VxFlexOS_NOVA/4ff1971/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"25d0557d7db7c5c282bfcc1f920d5d1d2629f83f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-13 02:12:08.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/8 : FAILURE in 2h 11m 44s","accounts_in_message":[],"_revision_number":8},{"id":"fad8a730da4268a9b5ebeba85dc2cd1173722a70","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-13 13:19:26.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"7848ac6c3169092d52318b36aa1d62d906f27e04","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 11:48:25.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/8/check/check-dsvm-tempest-vz7-exe-minimal/c25557e : FAILURE in 1h 14m 24s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/8/check/check-dsvm-tempest-vz7-exe-multinode/b87da12 : FAILURE in 38m 58s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":8},{"id":"0292932aea0257eb117f69806c45962fff441cbf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-17 13:21:24.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"61cb0b8fce464b03a256ce10d4aad8e6f98286c4","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-17 17:40:25.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"dd8c5a7adfe443eba9aba538fa393e4f658f7906","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 17:41:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5e075b274976f3b1fc3cd1405f6acdaa8ee5c400","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-17 19:01:37.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30638 : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":9},{"id":"518fa0b1e3f536ab3407d23f1985907fa77deeac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-17 19:03:45.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/9/check/tempest-dsvm-full-xenial/7b43128/ : SUCCESS in 1h 15m 19s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/9/check/tempest-dsvm-full-xenial-py3/07251fb/ : SUCCESS in 1h 22m 00s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/9/check/grenade-dsvm-xenial/a4077ce/ : SUCCESS in 1h 04m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"887c8f0f28b4761a9e6025ba59c91263551e6bd9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-17 20:00:38.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/9 : FAILURE in 2h 18m 38s","accounts_in_message":[],"_revision_number":9},{"id":"9d1aca862c556210b533ad89ae5ab5de5c7f3430","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 21:59:12.000000000","message":"Patch Set 9:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/9/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/6d291d1 : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":9},{"id":"63a1582ced8e696daa47152b491a21ef5cabd91d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 06:02:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/9/check/grenade-py3/4ab86b1/ : SUCCESS in 1h 02m 12s\n- tempest-full-py3 http://logs.openstack.org/96/639396/9/check/tempest-full-py3/0397006/ : SUCCESS in 1h 40m 39s\n- openstack-tox-cover http://logs.openstack.org/96/639396/9/check/openstack-tox-cover/0ab33ff/cover/ : SUCCESS in 18m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/9/check/openstack-tox-lower-constraints/368ff3b/ : SUCCESS in 18m 03s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/9/check/openstack-tox-pep8/e5fa10d/ : SUCCESS in 11m 55s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/9/check/openstack-tox-py27/0ca2608/ : SUCCESS in 13m 25s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/9/check/openstack-tox-py36/d0baa02/ : SUCCESS in 12m 57s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/9/check/openstack-tox-py37/9f30f3d/ : SUCCESS in 12m 13s\n- openstack-tox-docs http://logs.openstack.org/96/639396/9/check/openstack-tox-docs/0e7b3a0/html/ : SUCCESS in 6m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/9/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/33ef2e7/ : SUCCESS in 52m 51s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/9/check/devstack-plugin-ceph-tempest/21ab455/ : SUCCESS in 1h 39m 19s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/9/check/neutron-grenade-multinode/2021645/ : SUCCESS in 1h 05m 11s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/9/check/nova-grenade-live-migration/5df508d/ : SUCCESS in 1h 09m 21s\n- nova-live-migration http://logs.openstack.org/96/639396/9/check/nova-live-migration/3a811bd/ : SUCCESS in 54m 34s\n- nova-lvm http://logs.openstack.org/96/639396/9/check/nova-lvm/53793eb/ : SUCCESS in 1h 07m 44s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/9/check/nova-next/3c5d824/ : SUCCESS in 1h 32m 57s\n- nova-tox-functional http://logs.openstack.org/96/639396/9/check/nova-tox-functional/8f263b7/ : SUCCESS in 21m 02s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/9/check/nova-tox-functional-py36/18f0aae/ : SUCCESS in 17m 32s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/9/check/tempest-slow-py3/7eb3f47/ : SUCCESS in 2h 22m 06s","accounts_in_message":[],"_revision_number":9},{"id":"f976cde84b1356123885c2abd19a9d00dcea41c9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-18 06:13:38.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/9/check/EMC_VxFlexOS_NOVA/38e736b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"9cfbc94b802b222877618c70aca149387be121b7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-18 11:58:02.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"431ebb9832cade72a8f4aab5b2b798b1aa55c393","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-19 09:30:34.000000000","message":"Patch Set 9:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/9/check/nova-out-of-tree-pvm/d4456c3 : FAILURE in 2m 08s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/9/check/nova-in-tree-pvm/68137df : FAILURE in 2m 04s","accounts_in_message":[],"_revision_number":9},{"id":"804c52de8075bb783380690e7ea0c7fffd79615a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-23 17:43:04.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"9ac2b36f1f54f8e2faa116362a10796513b86fdc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-23 17:44:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b651b3fc8ac092c694d3e1e347d1bf28e2abd261","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-23 17:44:41.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/10/check/tempest-dsvm-full-xenial/8f3c6ed/ : FAILURE in 9s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/10/check/tempest-dsvm-full-xenial-py3/5e8fe8b/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/10/check/grenade-dsvm-xenial/c237afd/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"991c73e5de475ca5a4d12459f0eefc179ca087c3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-23 17:44:48.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30951 : FAILURE in 7s","accounts_in_message":[],"_revision_number":10},{"id":"4f6df20db906fb86c893e1b8702b8d3f316a87d8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-23 17:50:50.000000000","message":"Patch Set 10:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/0ea3894 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":10},{"id":"b1688bc8fe0f2f63314e1caa76c69953f90c4dfa","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-23 18:11:14.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/10/check/nova-out-of-tree-pvm/ed2410f : FAILURE in 1m 12s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/10/check/nova-in-tree-pvm/00816bb : FAILURE in 1m 18s","accounts_in_message":[],"_revision_number":10},{"id":"c4f8dea41f1c6cadba0bae934bc584d0469da9b1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-23 20:03:21.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/10 : SUCCESS in 2h 18m 57s","accounts_in_message":[],"_revision_number":10},{"id":"fd6d21e0a7a90ac2f358936710b4bb91a11a2d83","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 22:42:00.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/10/check-vote/ext-nova-zuul/ea748b2 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":10},{"id":"62ff98021fc25747d3d85734a7217b036460a563","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:45:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/10/check/grenade-py3/6a32af2/ : SUCCESS in 51m 52s\n- tempest-full-py3 http://logs.openstack.org/96/639396/10/check/tempest-full-py3/5c30e9c/ : SUCCESS in 1h 44m 08s\n- openstack-tox-cover http://logs.openstack.org/96/639396/10/check/openstack-tox-cover/20249c0/cover/ : SUCCESS in 16m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/10/check/openstack-tox-lower-constraints/f27e0c1/ : SUCCESS in 14m 34s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/10/check/openstack-tox-pep8/4b0776a/ : SUCCESS in 10m 17s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/10/check/openstack-tox-py27/b440cbd/ : SUCCESS in 12m 41s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/10/check/openstack-tox-py36/e6bb5ff/ : SUCCESS in 12m 23s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/10/check/openstack-tox-py37/3f5925f/ : SUCCESS in 12m 40s\n- openstack-tox-docs http://logs.openstack.org/96/639396/10/check/openstack-tox-docs/ad46977/html/ : SUCCESS in 6m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5d2f5a4/ : SUCCESS in 51m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/10/check/devstack-plugin-ceph-tempest/517bb08/ : SUCCESS in 1h 30m 26s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/10/check/neutron-grenade-multinode/f6648d8/ : SUCCESS in 1h 14m 13s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/10/check/nova-grenade-live-migration/dc874e2/ : SUCCESS in 1h 04m 58s\n- nova-live-migration nova-live-migration : ERROR Unable to find playbook /var/lib/zuul/builds/b6d25b7ec2354b34a4fa54ed3df8313f/untrusted/project_3/opendev.org/openstack/nova/playbooks/legacy/nova-live-migration/run.yaml\n- nova-lvm http://logs.openstack.org/96/639396/10/check/nova-lvm/6928247/ : SUCCESS in 57m 28s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/10/check/nova-next/a46ad87/ : SUCCESS in 1h 35m 46s\n- nova-tox-functional http://logs.openstack.org/96/639396/10/check/nova-tox-functional/bc7c596/ : SUCCESS in 18m 50s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/10/check/nova-tox-functional-py36/9b4ba39/ : SUCCESS in 18m 25s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/10/check/tempest-slow-py3/671a9ed/ : SUCCESS in 2h 03m 59s","accounts_in_message":[],"_revision_number":10},{"id":"2483ebe29ea5bbb14d8fa0170ae3b14cb25793fb","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-24 10:07:40.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"8a694991cfbf4760c361dfc2420673f6bfeef84d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:08:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4e4f9dc0f8fd48172443a660e6e3587dc745722f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 10:09:28.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30993 : FAILURE in 8s","accounts_in_message":[],"_revision_number":11},{"id":"b181148877e16b8429565c1ec1253098f0a34b49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 10:10:59.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/11/check/tempest-dsvm-full-xenial/e7926f0/ : FAILURE in 9s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/11/check/tempest-dsvm-full-xenial-py3/550f68e/ : FAILURE in 9s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/11/check/grenade-dsvm-xenial/e5ede02/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"eb11b466d485dd19f2889b45bbc35090f0809493","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:11:52.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/11/check/check-dsvm-tempest-vz7-exe-minimal/ae1d9e0 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/11/check/check-dsvm-tempest-vz7-exe-multinode/f70297d : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":11},{"id":"16c6810dc7f7cfe32d895d9af986e4e4a9a69df6","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:16:03.000000000","message":"Patch Set 11:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/4d75b43 : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":11},{"id":"ae754de5975923ea2d84af718ff5ef2f2ed455b8","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-24 10:18:36.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"48a91eaeb5041431c3f02f5d27d1c7632228be88","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:19:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4da74b1ecffa69ae4869751488249e15d78a3b48","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 10:21:41.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/12/check/tempest-dsvm-full-xenial/5cea30e/ : FAILURE in 14s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/12/check/tempest-dsvm-full-xenial-py3/74b28c0/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/12/check/grenade-dsvm-xenial/2e40c30/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ac1f6f78e908cdd8b7845bd89af2e37232b7d988","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:23:14.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/12/check/check-dsvm-tempest-vz7-exe-minimal/832cc06 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/12/check/check-dsvm-tempest-vz7-exe-multinode/bd4235b : FAILURE in 52s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":12},{"id":"74d01b39adf205ce27dee88bdc78a53c416df719","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 10:23:59.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30996 : FAILURE in 7s","accounts_in_message":[],"_revision_number":12},{"id":"f33ccde28efc5364320541c86c00f5f5b444bbc9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 10:27:24.000000000","message":"Patch Set 12:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/690bdf6 : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":12},{"id":"67e30a9ae60e03b5ad9940ef250dae16d604703f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-24 11:32:50.000000000","message":"Patch Set 12:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/12/check/nova-out-of-tree-pvm/f318e6b : FAILURE in 9m 15s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/12/check/nova-in-tree-pvm/1e88697 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":12},{"id":"072a27b4c57d97d6b4dfcc1f80eb481c273de3df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 13:34:24.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/12/check-vote/ext-nova-zuul/8abc5c6 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":12},{"id":"8b159bba4e1bb790639862eabbeb506fc7b8a69c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-24 13:35:27.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639396/12 : FAILURE in 2h 23m 00s","accounts_in_message":[],"_revision_number":12},{"id":"1f6835e2f916f5f2b7b906e49232dbbaf69174e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 16:16:18.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/12/check/grenade-py3/147f5af/ : SUCCESS in 57m 04s\n- tempest-full-py3 http://logs.openstack.org/96/639396/12/check/tempest-full-py3/9e2a65c/ : SUCCESS in 1h 55m 08s\n- openstack-tox-cover http://logs.openstack.org/96/639396/12/check/openstack-tox-cover/ec9df59/cover/ : SUCCESS in 20m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/12/check/openstack-tox-lower-constraints/8a55aa5/ : SUCCESS in 20m 52s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/12/check/openstack-tox-pep8/17bd77a/ : SUCCESS in 10m 04s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/12/check/openstack-tox-py27/0382e4c/ : SUCCESS in 12m 40s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/12/check/openstack-tox-py36/314bbb5/ : FAILURE in 11m 18s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/12/check/openstack-tox-py37/88c495f/ : SUCCESS in 11m 38s\n- openstack-tox-docs http://logs.openstack.org/96/639396/12/check/openstack-tox-docs/5911d3d/html/ : SUCCESS in 6m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/42e2036/ : SUCCESS in 50m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/12/check/devstack-plugin-ceph-tempest/44db183/ : SUCCESS in 1h 33m 53s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/12/check/neutron-grenade-multinode/08403ff/ : SUCCESS in 1h 07m 42s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/12/check/nova-grenade-live-migration/4a60a23/ : SUCCESS in 1h 03m 16s\n- nova-lvm http://logs.openstack.org/96/639396/12/check/nova-lvm/fe67217/ : SUCCESS in 54m 26s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/12/check/nova-next/701379f/ : SUCCESS in 1h 22m 09s\n- nova-tox-functional http://logs.openstack.org/96/639396/12/check/nova-tox-functional/d66a93f/ : SUCCESS in 22m 42s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/12/check/nova-tox-functional-py36/b030894/ : SUCCESS in 15m 48s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/12/check/tempest-slow-py3/8cced95/ : SUCCESS in 2h 28m 07s","accounts_in_message":[],"_revision_number":12},{"id":"9f4cf62a5cd1736392f72a508d99e4e3da1fc107","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-24 17:13:00.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"64f2f980312804589c092ed82796f3cca0c6a80f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 17:14:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ad36c8ca22311ca472ed93510a853a4464b66bd6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 17:14:44.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31012 : FAILURE in 7s","accounts_in_message":[],"_revision_number":13},{"id":"9efd34ad3ef7e37a1a273a443edbeca868dbf1a8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 17:15:18.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/13/check/tempest-dsvm-full-xenial/17c79f8/ : FAILURE in 9s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/13/check/tempest-dsvm-full-xenial-py3/36c4d8c/ : FAILURE in 9s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/13/check/grenade-dsvm-xenial/28d696a/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bc6cbb26c2ac73c4727f1569de9b4d57059ca306","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 17:16:59.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/13/check/check-dsvm-tempest-vz7-exe-minimal/37eeb40 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/13/check/check-dsvm-tempest-vz7-exe-multinode/34a45a4 : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":13},{"id":"e33e6f7f4162e54c23d2bb7f8e2e26a2c289a199","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 17:20:58.000000000","message":"Patch Set 13:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/6b2a460 : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":13},{"id":"eea935eb53b946e23d47d430ee01fa652ee252e9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-24 17:23:30.000000000","message":"Patch Set 13:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/13/check/nova-out-of-tree-pvm/baa3f6b : FAILURE in 9m 25s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/13/check/nova-in-tree-pvm/1f5b6f4 : FAILURE in 9m 42s","accounts_in_message":[],"_revision_number":13},{"id":"56d091167a621e3e2b5333a78b66fa8a74c8b527","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-24 17:51:47.000000000","message":"Patch Set 13:\n\n* pci-test http://52.27.155.124/pci/639396/13 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"8c3c0ac2007e67ff1dee40cb85e3f11b75228067","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-24 21:28:37.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/13 : SUCCESS in 2h 20m 10s","accounts_in_message":[],"_revision_number":13},{"id":"440dea390899c03a6576dfdbf467e6cdaa1c8c36","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 23:19:18.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/13/check-vote/ext-nova-zuul/fd9eaf6 : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":13},{"id":"ed152504d32a6f24f780f57f52527c50b5864bc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 05:33:55.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/13/check/grenade-py3/da1cb7d/ : SUCCESS in 58m 12s\n- tempest-full-py3 http://logs.openstack.org/96/639396/13/check/tempest-full-py3/81c34fb/ : SUCCESS in 1h 26m 32s\n- openstack-tox-cover http://logs.openstack.org/96/639396/13/check/openstack-tox-cover/977f12a/cover/ : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/13/check/openstack-tox-lower-constraints/ac0cd42/ : SUCCESS in 13m 54s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/13/check/openstack-tox-pep8/c10f38c/ : SUCCESS in 9m 48s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/13/check/openstack-tox-py27/8711a9c/ : SUCCESS in 14m 51s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/13/check/openstack-tox-py36/ffcc468/ : FAILURE in 12m 31s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/13/check/openstack-tox-py37/5760dd4/ : SUCCESS in 11m 02s\n- openstack-tox-docs http://logs.openstack.org/96/639396/13/check/openstack-tox-docs/56a9248/html/ : SUCCESS in 6m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/13/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b07fee8/ : SUCCESS in 52m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/13/check/devstack-plugin-ceph-tempest/343f5c0/ : SUCCESS in 1h 31m 00s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/13/check/neutron-grenade-multinode/707a7e3/ : SUCCESS in 1h 10m 49s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/13/check/nova-grenade-live-migration/7c1aebc/ : SUCCESS in 1h 08m 49s\n- nova-multinode http://logs.openstack.org/96/639396/13/check/nova-multinode/7d2ea95/ : SUCCESS in 55m 31s\n- nova-lvm http://logs.openstack.org/96/639396/13/check/nova-lvm/fccabb7/ : SUCCESS in 1h 06m 50s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/13/check/nova-next/b3bdafa/ : FAILURE in 1h 33m 52s\n- nova-tox-functional http://logs.openstack.org/96/639396/13/check/nova-tox-functional/6d76aa6/ : SUCCESS in 17m 39s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/13/check/nova-tox-functional-py36/1916253/ : SUCCESS in 20m 36s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/13/check/tempest-slow-py3/2d27af5/ : FAILURE in 2h 14m 35s","accounts_in_message":[],"_revision_number":13},{"id":"48b6f0097df5f0ea5b75f54d816d74ef138a3e96","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-25 17:51:17.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"cff1e2df7413e9a667514ea9c0fa8429e87bf164","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-25 17:52:32.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d62c70eb18db56ffbb18a7c76b5013cb9bf1a2ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-25 17:52:57.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/14/check/tempest-dsvm-full-xenial/264d437/ : FAILURE in 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/14/check/tempest-dsvm-full-xenial-py3/3c706a0/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/14/check/grenade-dsvm-xenial/cde39b3/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a18c16d5cac0e6b0c152d50bd3cc63a7d4b5ee83","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-25 17:53:05.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31060 : FAILURE in 8s","accounts_in_message":[],"_revision_number":14},{"id":"4c5e6477c63f5ba059c554341067dea4798b7b73","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-25 17:56:26.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/14/check/check-dsvm-tempest-vz7-exe-minimal/342a161 : FAILURE in 52s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/14/check/check-dsvm-tempest-vz7-exe-multinode/ba06cac : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":14},{"id":"efc8d6f0e41b0c820aa52de63a04fcf1ddb0aac4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-25 17:59:17.000000000","message":"Patch Set 14:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/14/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/bb03424 : FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":14},{"id":"2d30d96624f6ab67c1226c99934c9b7265f3f096","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-25 18:41:39.000000000","message":"Patch Set 14:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/14/check/nova-out-of-tree-pvm/6efe9c3 : FAILURE in 9m 37s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/14/check/nova-in-tree-pvm/f8cc9c3 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":14},{"id":"be6db97d3c8368e91a8d0edd02a0803e56cc230c","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-25 21:22:48.000000000","message":"Patch Set 14:\n\n* pci-test http://52.27.155.124/pci/639396/14 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"5453c4e78ff8e475d12525e843571fbcc9b4c591","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-25 22:42:14.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/14 : SUCCESS in 2h 13m 39s","accounts_in_message":[],"_revision_number":14},{"id":"eafa98e7cb999867da8bfe61ab0a82ac0d205295","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-25 23:08:28.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/14/check-vote/ext-nova-zuul/8e1111b : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":14},{"id":"6a1d71059c28a888c40ed7f7474f48615ae0b5a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:41:41.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/96/639396/14/check/grenade-py3/7380e61/ : SUCCESS in 52m 29s\n- tempest-full-py3 http://logs.openstack.org/96/639396/14/check/tempest-full-py3/103a8ea/ : SUCCESS in 1h 46m 53s\n- openstack-tox-cover http://logs.openstack.org/96/639396/14/check/openstack-tox-cover/bc3f1e4/cover/ : SUCCESS in 14m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/14/check/openstack-tox-lower-constraints/27be4c9/ : SUCCESS in 14m 00s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/14/check/openstack-tox-pep8/d6c869f/ : SUCCESS in 10m 17s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/14/check/openstack-tox-py27/adc72b2/ : SUCCESS in 11m 14s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/14/check/openstack-tox-py36/aa381c6/ : SUCCESS in 11m 27s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/14/check/openstack-tox-py37/d010581/ : SUCCESS in 17m 40s\n- openstack-tox-docs http://logs.openstack.org/96/639396/14/check/openstack-tox-docs/5e8a16d/html/ : SUCCESS in 6m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/14/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/adb446b/ : SUCCESS in 50m 51s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/14/check/devstack-plugin-ceph-tempest/1a0431c/ : SUCCESS in 1h 39m 10s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/14/check/neutron-grenade-multinode/04b8b20/ : SUCCESS in 1h 13m 27s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/14/check/nova-grenade-live-migration/f9ad207/ : SUCCESS in 1h 02m 46s\n- nova-multinode http://logs.openstack.org/96/639396/14/check/nova-multinode/82d7fc2/ : FAILURE in 2h 20m 14s\n- nova-lvm http://logs.openstack.org/96/639396/14/check/nova-lvm/1a03727/ : SUCCESS in 1h 07m 52s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/14/check/nova-next/7a9e061/ : SUCCESS in 1h 38m 12s\n- nova-tox-functional http://logs.openstack.org/96/639396/14/check/nova-tox-functional/aed1ac2/ : SUCCESS in 18m 34s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/14/check/nova-tox-functional-py36/f1cea24/ : SUCCESS in 19m 21s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/14/check/tempest-slow-py3/60fd531/ : SUCCESS in 2h 12m 45s","accounts_in_message":[],"_revision_number":14},{"id":"66abf456f7c619a381264911c9bcca74fd8b4c2a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-26 20:14:27.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"7340e1d27785a57e0812cbc81276968af7305e40","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-26 20:15:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d2a85cb8c406822b53870476522e4e3c9ea5dcb5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-26 20:16:41.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/15/check/tempest-dsvm-full-xenial/21908f3/ : FAILURE in 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/15/check/tempest-dsvm-full-xenial-py3/c36b32f/ : FAILURE in 10s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/15/check/grenade-dsvm-xenial/6592489/ : FAILURE in 9s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ba63244cb988a275c807baec52a8501bad8ba38f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-26 20:21:05.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31122 : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":15},{"id":"d71350525890026bfc988aa6a609f773b49959c6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-26 20:23:10.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/15/check/check-dsvm-tempest-vz7-exe-minimal/7446db6 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/15/check/check-dsvm-tempest-vz7-exe-multinode/87e960c : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"c051659a3b2621f873bb6498b11eead694c45634","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-26 20:25:10.000000000","message":"Patch Set 15:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/15/check/nova-out-of-tree-pvm/562e76f : FAILURE in 9m 13s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/15/check/nova-in-tree-pvm/8d87765 : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":15},{"id":"bd1c04904348350d8972dde92ab67bf5f6fc176e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-26 20:32:24.000000000","message":"Patch Set 15:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1049e44 : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":15},{"id":"4764b005dda40495a75cd6cc17658ff310998e1f","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-26 22:08:50.000000000","message":"Patch Set 15:\n\n* pci-test http://52.27.155.124/pci/639396/15 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"9681dcb7d854628d213021ef84584c5e7cd9f1f6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-27 01:40:11.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/15 : SUCCESS in 2h 26m 24s","accounts_in_message":[],"_revision_number":15},{"id":"97efd73492bb72c1095600983d5ebd88f9d6b7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 02:20:36.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/15/check/grenade-py3/770bf8c/ : SUCCESS in 1h 01m 01s\n- tempest-full-py3 http://logs.openstack.org/96/639396/15/check/tempest-full-py3/323224b/ : SUCCESS in 1h 36m 00s\n- openstack-tox-cover http://logs.openstack.org/96/639396/15/check/openstack-tox-cover/92cf0b2/cover/ : SUCCESS in 16m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/15/check/openstack-tox-lower-constraints/25c3645/ : SUCCESS in 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/15/check/openstack-tox-pep8/4baf6f7/ : SUCCESS in 10m 24s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/15/check/openstack-tox-py27/a2955f6/ : SUCCESS in 12m 48s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/15/check/openstack-tox-py36/c7923e4/ : SUCCESS in 13m 04s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/15/check/openstack-tox-py37/b1263bc/ : SUCCESS in 11m 09s\n- openstack-tox-docs http://logs.openstack.org/96/639396/15/check/openstack-tox-docs/df00f01/html/ : SUCCESS in 6m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/15/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9aef3f5/ : SUCCESS in 45m 35s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/15/check/devstack-plugin-ceph-tempest/c22db9e/ : SUCCESS in 1h 39m 25s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/15/check/neutron-grenade-multinode/9293980/ : SUCCESS in 1h 08m 57s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/15/check/nova-grenade-live-migration/9eb176a/ : SUCCESS in 1h 12m 01s\n- nova-multinode http://logs.openstack.org/96/639396/15/check/nova-multinode/f71bc5d/ : SUCCESS in 48m 28s\n- nova-lvm http://logs.openstack.org/96/639396/15/check/nova-lvm/5961b9b/ : SUCCESS in 1h 09m 34s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/15/check/nova-next/083f247/ : SUCCESS in 1h 33m 23s\n- nova-tox-functional http://logs.openstack.org/96/639396/15/check/nova-tox-functional/847a06e/ : SUCCESS in 22m 09s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/15/check/nova-tox-functional-py36/859c6c9/ : SUCCESS in 18m 25s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/15/check/tempest-slow-py3/d0dbad2/ : SUCCESS in 2h 19m 52s","accounts_in_message":[],"_revision_number":15},{"id":"095cd92525370d981a13c36fe09711e468c05bad","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-27 02:21:19.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/15/check/EMC_VxFlexOS_NOVA/c70557c/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"904c8645f5b25ff0e868eae377697f2d2dc3f734","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-27 02:33:13.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/15/check-vote/ext-nova-zuul/20d62af : FAILURE in 56m 44s","accounts_in_message":[],"_revision_number":15},{"id":"130093665fb0b77fe6431e92d534ed7fafb68448","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-27 02:38:50.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"86a8d6acd9085ed0c9431bb4ef2f484ce97f6daa","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-04-27 15:07:15.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/96/639396/15/check-nova/Nova-ML2-Sriov/7e5f4b8 : SUCCESS in 1h 26m 15s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/96/639396/15/check-nova/Nova-MACVTAP-ML2-Sriov/6e0033f : SUCCESS in 1h 17m 15s (non-voting)\n- NVMe http://13.74.249.42/96/639396/15/check-nova/NVMe/972c24b : FAILURE in 12m 32s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":15},{"id":"6c76d3f5aab84a8b9cbdf4d30ea0c98c215cfcff","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-04-29 13:58:17.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"c8fdc08cb734cee9a42eaf38b1d8e46520a8d848","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 13:59:25.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7cb2a2ac35fd4064f05840fd914144f37dcc4845","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:04:44.000000000","message":"Patch Set 16:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/16/check/check-dsvm-tempest-vz7-exe-minimal/ad1e523 : FAILURE in 28s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/96/639396/16/check/check-dsvm-tempest-vz7-exe-multinode/bedb727 : FAILURE in 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"d9be68312c81a516fa8625687515977b2f953017","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 14:05:59.000000000","message":"Patch Set 16:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639396/16/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/be6271e : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":16},{"id":"b123e2fc93bc971720c546a9ebcdd4d4275a2cf2","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-29 14:19:47.000000000","message":"Patch Set 16:\n\n* pci-test http://52.27.155.124/pci/639396/16 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"aaaa48b2d3bee594aa5479fb38ddae6edecb351d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 14:53:48.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/96/639396/16/check-vote/ext-nova-zuul/838eb59 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":16},{"id":"a90c33761272d5cb44a1bcd195b016772e02ad30","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 15:17:16.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/16/check/tempest-dsvm-full-xenial/e1c481a/ : SUCCESS in 1h 08m 39s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/16/check/tempest-dsvm-full-xenial-py3/c5eb4c4/ : SUCCESS in 1h 17m 26s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/96/639396/16/check/grenade-dsvm-xenial/2d1eedc/ : SUCCESS in 1h 06m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ac7f8feefd996f37f109d037c5a2341e988834ed","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-29 15:54:01.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31231 : FAILURE in 1h 54m 23s","accounts_in_message":[],"_revision_number":16},{"id":"6fd15bd6a133adadf1e6ec691c686b3a833cdc02","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-29 16:16:04.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639396/16 : SUCCESS in 2h 16m 28s","accounts_in_message":[],"_revision_number":16},{"id":"c22a47651c5811390468b61a8b8e4044ee2af1d4","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-29 16:32:35.000000000","message":"Patch Set 16:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/96/639396/16/check/nova-out-of-tree-pvm/0676365 : FAILURE in 2h 12m 16s\n- nova-in-tree-pvm http://184.172.12.213/96/639396/16/check/nova-in-tree-pvm/966f7ab : FAILURE in 2h 32m 40s","accounts_in_message":[],"_revision_number":16},{"id":"4c7a357763d5b79b3798a71f5523af6ab3939bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:54:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/96/639396/16/check/grenade-py3/aaa198d/ : SUCCESS in 1h 06m 05s\n- tempest-full-py3 http://logs.openstack.org/96/639396/16/check/tempest-full-py3/92639d6/ : SUCCESS in 1h 23m 01s\n- openstack-tox-cover http://logs.openstack.org/96/639396/16/check/openstack-tox-cover/c5e9678/cover/ : SUCCESS in 14m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/639396/16/check/openstack-tox-lower-constraints/06c81fa/ : SUCCESS in 13m 54s\n- openstack-tox-pep8 http://logs.openstack.org/96/639396/16/check/openstack-tox-pep8/3e80cbe/ : SUCCESS in 10m 34s\n- openstack-tox-py27 http://logs.openstack.org/96/639396/16/check/openstack-tox-py27/435f6cc/ : SUCCESS in 14m 21s\n- openstack-tox-py36 http://logs.openstack.org/96/639396/16/check/openstack-tox-py36/8dbba5d/ : SUCCESS in 13m 43s\n- openstack-tox-py37 http://logs.openstack.org/96/639396/16/check/openstack-tox-py37/2821402/ : SUCCESS in 13m 18s\n- openstack-tox-docs http://logs.openstack.org/96/639396/16/check/openstack-tox-docs/7e2b831/html/ : SUCCESS in 6m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/96/639396/16/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6cb75f9/ : SUCCESS in 55m 49s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/639396/16/check/devstack-plugin-ceph-tempest/14d4f3f/ : SUCCESS in 1h 36m 20s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/639396/16/check/neutron-grenade-multinode/8ba1909/ : SUCCESS in 1h 14m 39s\n- nova-grenade-live-migration http://logs.openstack.org/96/639396/16/check/nova-grenade-live-migration/783e8e6/ : SUCCESS in 1h 13m 57s\n- nova-multinode http://logs.openstack.org/96/639396/16/check/nova-multinode/ee1e0b0/ : SUCCESS in 55m 08s\n- nova-lvm http://logs.openstack.org/96/639396/16/check/nova-lvm/ae81caf/ : SUCCESS in 59m 38s (non-voting)\n- nova-next http://logs.openstack.org/96/639396/16/check/nova-next/1707713/ : SUCCESS in 1h 33m 16s\n- nova-tox-functional http://logs.openstack.org/96/639396/16/check/nova-tox-functional/80edf63/ : SUCCESS in 17m 29s\n- nova-tox-functional-py36 http://logs.openstack.org/96/639396/16/check/nova-tox-functional-py36/392d84d/ : SUCCESS in 15m 51s\n- tempest-slow-py3 http://logs.openstack.org/96/639396/16/check/tempest-slow-py3/477875e/ : SUCCESS in 2h 16m 20s","accounts_in_message":[],"_revision_number":16},{"id":"339297d1c3c85c8f0897df52a83af4d0cbab7e13","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 16:55:10.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/96/639396/16/check/EMC_VxFlexOS_NOVA/2bd076d/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":16},{"id":"efefbadc00663fab79837f69dbf3e3c3f3bba53d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-29 18:04:50.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-96-639396-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"db53fcdc90651b3d30b3ccffdd2751da9f6146c7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-27 18:18:59.000000000","message":"Abandoned\n\nAbandoned, see commit message of [1] for explanation. Looks like we still need to wait in the virt driver, just not in all cases. Patch on top will be modified.\n\n[1] https://review.opendev.org/#/c/660782/1","accounts_in_message":[],"_revision_number":16},{"id":"5912650ab8f55149840a5c205f6dacfeff0a9cd0","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-27 18:20:19.000000000","message":"Restored","accounts_in_message":[],"_revision_number":16},{"id":"ed98c8716886b1086c33d5ffa314b23e94aced8b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-27 18:20:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"7a34b85d6971795c87a2ecc04a5000c461444815","revisions":{"0ca8c650bb57e7787952376240f43997886c7063":{"kind":"REWORK","_number":1,"created":"2019-02-26 17:37:50.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/1"}}},"commit":{"parents":[{"commit":"f7fd6cb58b22ad925f10a736215f4f004b497dc7","subject":"Merge \"quota: remove defaults kwarg in get_project_quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7fd6cb58b22ad925f10a736215f4f004b497dc7"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:06.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:38.000000000","tz":-300},"subject":"[Broken WIP] Split external events handling into 2 steps","message":"[Broken WIP] Split external events handling into 2 steps\n\nCloses-bug: 1813789\n\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ca8c650bb57e7787952376240f43997886c7063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ca8c650bb57e7787952376240f43997886c7063"}]},"branch":"refs/heads/master"},"33e6cda41535aa531304050d604ec09119c764ca":{"kind":"REWORK","_number":2,"created":"2019-03-07 20:11:46.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/2"}}},"commit":{"parents":[{"commit":"eaaade924d81ec597f30c6ca2f21c726dbbf2f6b","subject":"Merge \"Make nova-grenade-live-migration voting and gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaaade924d81ec597f30c6ca2f21c726dbbf2f6b"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:06.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-07 20:10:59.000000000","tz":-300},"subject":"Dont\u0027 wait for VIF plugging during revert resize","message":"Dont\u0027 wait for VIF plugging during revert resize\n\nCurrently, when reverting a resize, we call Neutron from the compute\nmanager to setup the instance networks on the source host. Later, in\nthe libvirt driver, we plugs the VIF(s)s and start the guest paused,\nthen wait for network-vif-plugged events from Neutron. The reasonning\nis that once Nova plugs the VIF(s), the Neutron agent wires them up on\nthe host. Once it\u0027s done, we can unpause the guest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch uses the existing vifs_already_plugged kwarg to\n_create_domain_and_network() to disable waiting for\nnetwork-vif-plugged events during a revert resize operation.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33e6cda41535aa531304050d604ec09119c764ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33e6cda41535aa531304050d604ec09119c764ca"}]},"branch":"refs/heads/master"},"2e55989a4873baf915cb006c4c19dbd0d2c2e429":{"kind":"REWORK","_number":3,"created":"2019-03-08 14:57:05.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/3"}}},"commit":{"parents":[{"commit":"eaaade924d81ec597f30c6ca2f21c726dbbf2f6b","subject":"Merge \"Make nova-grenade-live-migration voting and gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaaade924d81ec597f30c6ca2f21c726dbbf2f6b"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:06.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 14:55:57.000000000","tz":-300},"subject":"Dont\u0027 wait for VIF plugging during revert resize","message":"Dont\u0027 wait for VIF plugging during revert resize\n\nCurrently, when reverting a resize, we call Neutron from the compute\nmanager to setup the instance networks on the source host. Later, in\nthe libvirt driver, we plugs the VIF(s)s and start the guest paused,\nthen wait for network-vif-plugged events from Neutron. The reasonning\nis that once Nova plugs the VIF(s), the Neutron agent wires them up on\nthe host. Once it\u0027s done, we can unpause the guest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch uses the existing vifs_already_plugged kwarg to\n_create_domain_and_network() to disable waiting for\nnetwork-vif-plugged events during a revert resize operation.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e55989a4873baf915cb006c4c19dbd0d2c2e429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e55989a4873baf915cb006c4c19dbd0d2c2e429"}]},"branch":"refs/heads/master"},"bfb58980d1f593571c164eba84c1142faa3e0eaa":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-03-08 16:58:46.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/4"}}},"commit":{"parents":[{"commit":"eaaade924d81ec597f30c6ca2f21c726dbbf2f6b","subject":"Merge \"Make nova-grenade-live-migration voting and gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaaade924d81ec597f30c6ca2f21c726dbbf2f6b"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:06.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 16:58:08.000000000","tz":-300},"subject":"Dont\u0027 wait for VIF plugging during revert resize","message":"Dont\u0027 wait for VIF plugging during revert resize\n\nCurrently, when reverting a resize, we call Neutron from the compute\nmanager to setup the instance networks on the source host. Later, in\nthe libvirt driver, we plug the VIF(s)s and start the guest paused,\nthen wait for network-vif-plugged events from Neutron. The reasoning\nis that once Nova plugs the VIF(s), the Neutron agent wires them up on\nthe host. Once it\u0027s done, we can unpause the guest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch uses the existing vifs_already_plugged kwarg to\n_create_domain_and_network() to disable waiting for\nnetwork-vif-plugged events during a revert resize operation.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bfb58980d1f593571c164eba84c1142faa3e0eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bfb58980d1f593571c164eba84c1142faa3e0eaa"}]},"branch":"refs/heads/master"},"5ee4015d93f4f770aa34ec1b4b556b4f2fe904e0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-03-08 16:59:55.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/5"}}},"commit":{"parents":[{"commit":"eaaade924d81ec597f30c6ca2f21c726dbbf2f6b","subject":"Merge \"Make nova-grenade-live-migration voting and gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eaaade924d81ec597f30c6ca2f21c726dbbf2f6b"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-02-26 17:36:06.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 16:59:45.000000000","tz":-300},"subject":"Don\u0027t wait for VIF plugging during revert resize","message":"Don\u0027t wait for VIF plugging during revert resize\n\nCurrently, when reverting a resize, we call Neutron from the compute\nmanager to setup the instance networks on the source host. Later, in\nthe libvirt driver, we plug the VIF(s)s and start the guest paused,\nthen wait for network-vif-plugged events from Neutron. The reasoning\nis that once Nova plugs the VIF(s), the Neutron agent wires them up on\nthe host. Once it\u0027s done, we can unpause the guest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch uses the existing vifs_already_plugged kwarg to\n_create_domain_and_network() to disable waiting for\nnetwork-vif-plugged events during a revert resize operation.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ee4015d93f4f770aa34ec1b4b556b4f2fe904e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ee4015d93f4f770aa34ec1b4b556b4f2fe904e0"}]},"branch":"refs/heads/master"},"747ec2c483ca62a8a3072a4047852bcc1ce59e07":{"kind":"REWORK","_number":6,"created":"2019-03-20 14:46:57.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/6"}}},"commit":{"parents":[{"commit":"4f9bc724010f0c935bf83a6d19bdd805e86b7086","subject":"Merge \"doc: mark the max microversion for stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f9bc724010f0c935bf83a6d19bdd805e86b7086"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:48:01.000000000","tz":-300},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/747ec2c483ca62a8a3072a4047852bcc1ce59e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/747ec2c483ca62a8a3072a4047852bcc1ce59e07"}]},"branch":"refs/heads/master"},"08a480b0be7578f946e594d6c7efb419544764ab":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-04-10 23:51:47.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/7"}}},"commit":{"parents":[{"commit":"d42a007425d9adb691134137e1e0b7dda356df62","subject":"Merge \"Improve test coverage of nova.privsep.fs, continued.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d42a007425d9adb691134137e1e0b7dda356df62"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-10 23:51:31.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08a480b0be7578f946e594d6c7efb419544764ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08a480b0be7578f946e594d6c7efb419544764ab"}]},"branch":"refs/heads/master"},"f3ab31690aeae71b8d1f56f9a4012496c8929497":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-12 23:59:22.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/8"}}},"commit":{"parents":[{"commit":"cc7f09a81c9b42b1cec5c25016702a3d22c45417","subject":"DNM: extra logging in neutron api for revert resize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc7f09a81c9b42b1cec5c25016702a3d22c45417"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-12 23:59:08.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3ab31690aeae71b8d1f56f9a4012496c8929497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3ab31690aeae71b8d1f56f9a4012496c8929497"}]},"branch":"refs/heads/master"},"4b37e696ebb0e059e5dd7be6c412c3e70a1a47b8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-04-17 17:40:25.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/9"}}},"commit":{"parents":[{"commit":"380f22ec9262f6c0d16ada0e6a04b31b9380b57e","subject":"DNM: Run test_resize_server_revert in the live migration job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/380f22ec9262f6c0d16ada0e6a04b31b9380b57e"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-17 17:40:14.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b37e696ebb0e059e5dd7be6c412c3e70a1a47b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b37e696ebb0e059e5dd7be6c412c3e70a1a47b8"}]},"branch":"refs/heads/master"},"3acd6cf1cb6986e5605c355b0dc5d3124d0f0f3b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-04-23 17:43:04.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/10"}}},"commit":{"parents":[{"commit":"b676c65c56fab16b0e73f7398e9425e4c9f9266a","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b676c65c56fab16b0e73f7398e9425e4c9f9266a"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-23 17:42:47.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3acd6cf1cb6986e5605c355b0dc5d3124d0f0f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3acd6cf1cb6986e5605c355b0dc5d3124d0f0f3b"}]},"branch":"refs/heads/master"},"83925cd6fa7ca7098ed9bf78fa8fc0109547b0c0":{"kind":"NO_CHANGE","_number":11,"created":"2019-04-24 10:07:40.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/11"}}},"commit":{"parents":[{"commit":"1167fd2f39bbae56642d7af1b910fe94bf76f848","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1167fd2f39bbae56642d7af1b910fe94bf76f848"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-24 10:07:23.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83925cd6fa7ca7098ed9bf78fa8fc0109547b0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83925cd6fa7ca7098ed9bf78fa8fc0109547b0c0"}]},"branch":"refs/heads/master"},"e99975ea462cc3687980ff523d85f1e02fb0586e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-24 10:18:36.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/12"}}},"commit":{"parents":[{"commit":"44acb29a211cbf18ef0dea361fdb21c4b244b162","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44acb29a211cbf18ef0dea361fdb21c4b244b162"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-24 10:18:20.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e99975ea462cc3687980ff523d85f1e02fb0586e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e99975ea462cc3687980ff523d85f1e02fb0586e"}]},"branch":"refs/heads/master"},"2c803925ca7e10cf0396577b10d19d0fdadee168":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-04-24 17:13:00.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/13"}}},"commit":{"parents":[{"commit":"261a0812da3117f6f4533b9dd356c67e7de448e6","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/261a0812da3117f6f4533b9dd356c67e7de448e6"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-24 17:12:38.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c803925ca7e10cf0396577b10d19d0fdadee168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c803925ca7e10cf0396577b10d19d0fdadee168"}]},"branch":"refs/heads/master"},"dfdedd3fcba2e457286b596efc2f8ff62ae40526":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-25 17:51:17.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/14"}}},"commit":{"parents":[{"commit":"d9e32974d7a1b6c7d8c219387d06f7983726780a","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9e32974d7a1b6c7d8c219387d06f7983726780a"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-25 17:51:08.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dfdedd3fcba2e457286b596efc2f8ff62ae40526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dfdedd3fcba2e457286b596efc2f8ff62ae40526"}]},"branch":"refs/heads/master"},"b98ca4259e26c23e02a725fc13e0ad90a04957f6":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-04-26 20:14:27.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/15"}}},"commit":{"parents":[{"commit":"663da127b49f62a47f4b483b28d14ff0ffe4c9e4","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/663da127b49f62a47f4b483b28d14ff0ffe4c9e4"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-26 20:14:17.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b98ca4259e26c23e02a725fc13e0ad90a04957f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b98ca4259e26c23e02a725fc13e0ad90a04957f6"}]},"branch":"refs/heads/master"},"7a34b85d6971795c87a2ecc04a5000c461444815":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-29 13:58:17.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/639396/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/96/639396/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/96/639396/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/96/639396/16"}}},"commit":{"parents":[{"commit":"002e1242b8a716d14365df801312f7d6d4c910ba","subject":"Run revert resize tests in nova-live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/002e1242b8a716d14365df801312f7d6d4c910ba"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-03-08 20:43:58.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-04-29 13:58:02.000000000","tz":-240},"subject":"Revert \"Wait for network-vif-plugged on resize revert\"","message":"Revert \"Wait for network-vif-plugged on resize revert\"\n\nCurrently, when reverting a resize, we ask the Neutron API to\nmigrate_instance_finish(), which updates the port bindings back to the\nsource host. Later, in the libvirt driver, we plug the VIF(s)s and\nstart the guest paused, then wait for network-vif-plugged events from\nNeutron. The reasoning is that once Nova plugs the VIF(s), the Neutron\nagent wires them up on the host. Once it\u0027s done, we can unpause the\nguest.\n\nThis reasoning is false. On the source, the instance remains shut off,\nand the VIFs remain plugged. As soon as we call Neutron from the\ncompute manager, the agent is able to wire up the VIF(s) and we\nreceive the network-vif-plugged event. This races with the libvirt\ndriver setting up the event listener.\n\nThis patch reverts commit 4817165fc5938a553fafa1a69c6086f9ebe311af,\nwhich is when we started waiting for events during revert resize in\nthe libvirt driver. In a subsequent patch, the waiting will be done in\nthe compute manager instead.\n\nPartial-bug: 1813789\nChange-Id: I125e2af11ebcdc444c62ed41a9ce7985db6d5576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a34b85d6971795c87a2ecc04a5000c461444815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a34b85d6971795c87a2ecc04a5000c461444815"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
