)]}'
{"id":"openstack%2Fnova~639581","triplet_id":"openstack%2Fnova~master~I11bf62d5ef3041541459d138bedb50aaeb702ae2","project":"openstack/nova","branch":"master","topic":"my-own-personal-alternative-universe","hashtags":[],"change_id":"I11bf62d5ef3041541459d138bedb50aaeb702ae2","subject":"Cleanup the _execute shim in nova/network.","status":"MERGED","created":"2019-02-27 06:59:29.000000000","updated":"2019-08-22 08:13:00.000000000","submitted":"2019-08-22 08:12:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639581-1566461580428-8b5338a5","meta_rev_id":"f824cc9956bc509a23d40e8e42b96398222f6d5b","_number":639581,"virtual_id_number":639581,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 14:34:04.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 16:07:20.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 10:21:02.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 11:04:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-08-21 08:16:51.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-08-22 08:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"date":"2019-08-21 18:32:43.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-08-21 12:38:50.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"date":"2019-08-21 07:03:10.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":1,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-08-20 11:34:45.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-20 11:34:45.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":1,"date":"2019-08-20 11:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-20 14:48:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-01 02:48:32.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-03-06 19:09:40.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-03-10 13:12:31.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2019-03-11 12:24:46.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2019-03-21 17:47:35.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-03-26 20:11:39.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-27 09:47:25.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2019-04-13 11:44:25.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-06-26 20:31:25.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 05:03:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 09:07:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 12:21:52.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 14:34:04.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 14:48:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-08-20 16:07:20.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 07:03:10.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 08:16:51.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-08-21 10:21:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 11:04:45.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 12:38:50.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-21 18:32:43.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 08:12:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b06fa06ce12d3fb908c55b6efbc4a77e4806e414","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-02-27 06:59:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d33c139d376ba5cf71affa8dab424e982dc67b1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 07:11:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c67b7be4e4b2457cfbca92b38707024b65fbaf2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-02-27 08:23:31.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28317 : FAILURE in 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ff5ebf933efab5e82764681b370a9cd61a0a4f3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 09:36:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/1/check/grenade-py3/7899d54/ : SUCCESS in 57m 27s\n- tempest-full-py3 http://logs.openstack.org/81/639581/1/check/tempest-full-py3/c9a3e77/ : SUCCESS in 1h 45m 59s\n- openstack-tox-cover http://logs.openstack.org/81/639581/1/check/openstack-tox-cover/9892f85/cover/ : SUCCESS in 20m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/1/check/openstack-tox-lower-constraints/43895a8/ : SUCCESS in 13m 00s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/1/check/openstack-tox-pep8/4b441b7/ : SUCCESS in 10m 45s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/1/check/openstack-tox-py27/794a978/ : SUCCESS in 12m 44s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/1/check/openstack-tox-py35/3ddf028/ : SUCCESS in 13m 57s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/1/check/openstack-tox-py36/44ebfd2/ : SUCCESS in 12m 55s\n- openstack-tox-docs http://logs.openstack.org/81/639581/1/check/openstack-tox-docs/955651d/html/ : SUCCESS in 7m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/303eb53/ : SUCCESS in 48m 00s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/1/check/devstack-plugin-ceph-tempest/1d5726b/ : TIMED_OUT in 2h 04m 37s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639581/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/ce6686e/ : SUCCESS in 1h 07m 12s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/1/check/neutron-grenade-multinode/cd83b43/ : SUCCESS in 1h 16m 02s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/1/check/neutron-tempest-linuxbridge/ce4a497/ : SUCCESS in 1h 24m 17s\n- nova-live-migration http://logs.openstack.org/81/639581/1/check/nova-live-migration/cb46599/ : SUCCESS in 44m 06s\n- nova-next http://logs.openstack.org/81/639581/1/check/nova-next/a57c1d5/ : SUCCESS in 1h 55m 23s\n- nova-tox-functional http://logs.openstack.org/81/639581/1/check/nova-tox-functional/6652f27/ : SUCCESS in 21m 39s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/1/check/nova-tox-functional-py35/71feed0/ : SUCCESS in 21m 05s\n- tempest-multinode-full http://logs.openstack.org/81/639581/1/check/tempest-multinode-full/06bfa79/ : SUCCESS in 1h 42m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/1/check/tempest-slow-py3/b40b571/ : SUCCESS in 2h 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"039df4f9a74c744912ea90cef14fafabf2c579ae","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 09:40:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/1/check/EMC_VxFlexOS_NOVA/8e4643e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"dea9132b901eeeec934e2cabe268b9626a44cbbc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-27 13:17:40.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"f469e80bef97b576d2a8e9ba610ff63e0898855a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-02-27 23:04:26.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/1/check-vote/ext-nova-zuul/b585bc2 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0dfa93d0dcf39c84cecc18378946ac183846d80c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 02:26:00.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/0074d92 : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4285b6fc481743008acadd948693105cf032be7c","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-01 02:48:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"11ccacbfbe98898ddd866cebc3b6731a2c83d759","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-01 10:22:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/1/check/tempest-dsvm-full-xenial/a326c30/ : FAILURE in 9m 43s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/1/check/tempest-dsvm-full-xenial-py3/4cb6fe1/ : FAILURE in 6m 23s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/1/check/grenade-dsvm-xenial/b9ed984/ : RETRY_LIMIT (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9194a874ff9aa14e754c2987570ed668d546173","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-05 22:35:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"346bb332067935cc0ce8c81efc8252cdaf41bad9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-05 23:20:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c28929a390da01f91b69e0dd260d168fac10d2fa","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-06 00:30:01.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/2/check/nova-out-of-tree-pvm/8fa18c5 : FAILURE in 45m 06s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/2/check/nova-in-tree-pvm/a04f9a4 : FAILURE in 44m 06s","accounts_in_message":[],"_revision_number":2},{"id":"49e1572b2211f5e0da187c3c814b504b5672c8e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 04:09:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/2/check/grenade-py3/e59ddca/ : SUCCESS in 59m 47s\n- tempest-full-py3 http://logs.openstack.org/81/639581/2/check/tempest-full-py3/788c7f1/ : SUCCESS in 1h 19m 41s\n- openstack-tox-cover http://logs.openstack.org/81/639581/2/check/openstack-tox-cover/bf16b6d/cover/ : SUCCESS in 20m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/2/check/openstack-tox-lower-constraints/ec1b426/ : SUCCESS in 12m 51s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/2/check/openstack-tox-pep8/20ee9a9/ : SUCCESS in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/2/check/openstack-tox-py27/ab60321/ : SUCCESS in 15m 40s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/2/check/openstack-tox-py35/fec5942/ : SUCCESS in 14m 32s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/2/check/openstack-tox-py36/4933bb6/ : SUCCESS in 12m 44s\n- openstack-tox-docs http://logs.openstack.org/81/639581/2/check/openstack-tox-docs/a10a4ef/html/ : SUCCESS in 6m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ac2ff93/ : SUCCESS in 54m 02s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/2/check/devstack-plugin-ceph-tempest/ce228d7/ : SUCCESS in 1h 17m 10s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639581/2/check/legacy-grenade-dsvm-neutron-multinode-live-migration/a6bf7c3/ : SUCCESS in 1h 10m 49s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/2/check/neutron-grenade-multinode/e66c0e5/ : SUCCESS in 1h 10m 26s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/2/check/neutron-tempest-linuxbridge/4ef0b8c/ : SUCCESS in 1h 33m 05s\n- nova-live-migration http://logs.openstack.org/81/639581/2/check/nova-live-migration/af2421f/ : SUCCESS in 49m 12s\n- nova-next http://logs.openstack.org/81/639581/2/check/nova-next/5290478/ : SUCCESS in 2h 17m 47s\n- nova-tox-functional http://logs.openstack.org/81/639581/2/check/nova-tox-functional/5aaab3e/ : SUCCESS in 21m 31s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/2/check/nova-tox-functional-py35/4fdafc4/ : SUCCESS in 19m 08s\n- tempest-multinode-full http://logs.openstack.org/81/639581/2/check/tempest-multinode-full/bd10123/ : SUCCESS in 1h 55m 05s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/2/check/tempest-slow-py3/b51e319/ : SUCCESS in 2h 27m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f870db7836c140166d954fede2787a2bcf439a05","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-06 04:12:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/2/check/EMC_VxFlexOS_NOVA/e69abfa/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"bc78306bb9251758ddb6b44c650fdb2e2ac1ff0e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 05:29:30.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28742 : SUCCESS in 2h 24m 08s","accounts_in_message":[],"_revision_number":2},{"id":"5a0d68dfaa1d2448ef872321665016dce27233cf","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-06 06:08:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/2 : FAILURE in 2h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6acbe1e49b000d3257461ea6d95f8ffd5c915db5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-06 12:05:16.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"0059c23606a300cf2b7fb267c93f8668db1938eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 14:52:07.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/2/check-vote/ext-nova-zuul/b2e4765 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e2711c64a61bca1c06ed75e929599fbcadbc483f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-03-06 19:09:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c682e04e182fd263cd8c4332bfad92fbee89a8a8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 23:55:55.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28829 : SUCCESS in 1h 59m 06s","accounts_in_message":[],"_revision_number":3},{"id":"67544c29986041a38d98410d012f85f21242c8b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 00:34:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/3/check/grenade-py3/3856923/ : SUCCESS in 58m 16s\n- tempest-full-py3 http://logs.openstack.org/81/639581/3/check/tempest-full-py3/b70ecde/ : SUCCESS in 1h 32m 15s\n- openstack-tox-cover http://logs.openstack.org/81/639581/3/check/openstack-tox-cover/7b72627/cover/ : SUCCESS in 19m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/3/check/openstack-tox-lower-constraints/6ec51a7/ : SUCCESS in 13m 47s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/3/check/openstack-tox-pep8/0736fcd/ : SUCCESS in 11m 52s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/3/check/openstack-tox-py27/9af0759/ : SUCCESS in 15m 00s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/3/check/openstack-tox-py35/518cd48/ : SUCCESS in 13m 04s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/3/check/openstack-tox-py36/92f4750/ : SUCCESS in 13m 03s\n- openstack-tox-docs http://logs.openstack.org/81/639581/3/check/openstack-tox-docs/ba8d1dd/html/ : SUCCESS in 7m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/a255612/ : SUCCESS in 54m 01s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/3/check/devstack-plugin-ceph-tempest/ea26b74/ : SUCCESS in 1h 31m 22s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639581/3/check/legacy-grenade-dsvm-neutron-multinode-live-migration/0c815df/ : SUCCESS in 1h 08m 23s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/3/check/neutron-grenade-multinode/8287921/ : SUCCESS in 1h 14m 52s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/3/check/neutron-tempest-linuxbridge/c419190/ : SUCCESS in 1h 30m 21s\n- nova-live-migration http://logs.openstack.org/81/639581/3/check/nova-live-migration/970a50f/ : SUCCESS in 48m 46s\n- nova-next http://logs.openstack.org/81/639581/3/check/nova-next/3fa0384/ : SUCCESS in 1h 46m 15s\n- nova-tox-functional http://logs.openstack.org/81/639581/3/check/nova-tox-functional/cc22f81/ : SUCCESS in 21m 03s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/3/check/nova-tox-functional-py35/31d70f4/ : SUCCESS in 17m 55s\n- tempest-multinode-full http://logs.openstack.org/81/639581/3/check/tempest-multinode-full/f7889d9/ : SUCCESS in 1h 46m 38s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/3/check/tempest-slow-py3/a1e10ea/ : SUCCESS in 2h 23m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a6fcf02bd7367b6ecc62b6d300f0a3fdecb837d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-07 00:36:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/3/check/EMC_VxFlexOS_NOVA/2aef904/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"6e637574ebb1f45d6af99cd7b0b5857c193ddf91","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 00:59:17.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/3 : FAILURE in 41m 46s","accounts_in_message":[],"_revision_number":3},{"id":"89bcbdcc9da6465615b51c1768c98a007d4a65cb","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-07 12:06:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"776277b96e2a3b49db4dff665412b5133ef2c22c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 15:04:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/4/check/grenade-py3/45a8a5b/ : SUCCESS in 1h 01m 01s\n- tempest-full-py3 http://logs.openstack.org/81/639581/4/check/tempest-full-py3/7809961/ : SUCCESS in 1h 20m 17s\n- openstack-tox-cover http://logs.openstack.org/81/639581/4/check/openstack-tox-cover/5684a1e/ : FAILURE in 15m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/4/check/openstack-tox-lower-constraints/2ea4451/ : FAILURE in 11m 44s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/4/check/openstack-tox-pep8/f292049/ : FAILURE in 12m 21s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/4/check/openstack-tox-py27/19511d1/ : FAILURE in 12m 29s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/4/check/openstack-tox-py35/ea178db/ : FAILURE in 13m 04s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/4/check/openstack-tox-py36/9581b49/ : FAILURE in 12m 30s\n- openstack-tox-docs http://logs.openstack.org/81/639581/4/check/openstack-tox-docs/d7da98a/html/ : SUCCESS in 7m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/bfe7e6c/ : SUCCESS in 44m 01s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/4/check/devstack-plugin-ceph-tempest/820e9a8/ : SUCCESS in 1h 14m 06s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/81/639581/4/check/legacy-grenade-dsvm-neutron-multinode-live-migration/315575a/ : SUCCESS in 1h 07m 16s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/4/check/neutron-grenade-multinode/70e9df4/ : SUCCESS in 1h 12m 13s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/4/check/neutron-tempest-linuxbridge/af68ca3/ : SUCCESS in 1h 24m 59s\n- nova-live-migration http://logs.openstack.org/81/639581/4/check/nova-live-migration/18c2a6d/ : SUCCESS in 49m 18s\n- nova-next http://logs.openstack.org/81/639581/4/check/nova-next/82d6bc4/ : FAILURE in 1h 56m 34s\n- nova-tox-functional http://logs.openstack.org/81/639581/4/check/nova-tox-functional/bf47ed4/ : SUCCESS in 21m 18s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/4/check/nova-tox-functional-py35/de7635c/ : SUCCESS in 18m 52s\n- tempest-multinode-full http://logs.openstack.org/81/639581/4/check/tempest-multinode-full/573c127/ : SUCCESS in 1h 31m 07s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/4/check/tempest-slow-py3/4f25012/ : SUCCESS in 2h 17m 52s","accounts_in_message":[],"_revision_number":4},{"id":"520f5f58814320baa13e10a99333b77eaf5b3205","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-07 15:51:48.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28883 : SUCCESS in 2h 11m 53s","accounts_in_message":[],"_revision_number":4},{"id":"5d4c8aa9ee0ce39f3acd15bc9fa582d96215b662","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-07 17:11:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/4 : SUCCESS in 2h 30m 11s","accounts_in_message":[],"_revision_number":4},{"id":"7fa40249d147587520488501538ab5deaaa06f79","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-07 17:39:29.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"46f90a49ff95928e948f89395cf5092e2909b25e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-07 23:35:18.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/4/check-vote/ext-nova-zuul/5e03186 : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":4},{"id":"fa5a432c0937652ac4f0381f69ee46d96f3891bb","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-08 05:17:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3cbf4d2182f21f3ae111a1ec3631d8152ad622e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-08 07:35:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/5/check/grenade-py3/b5e9c77/ : SUCCESS in 1h 02m 55s\n- tempest-full-py3 http://logs.openstack.org/81/639581/5/check/tempest-full-py3/c0d3055/ : SUCCESS in 1h 39m 46s\n- openstack-tox-cover http://logs.openstack.org/81/639581/5/check/openstack-tox-cover/8e860f2/cover/ : SUCCESS in 15m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/5/check/openstack-tox-lower-constraints/e6e74e7/ : SUCCESS in 15m 04s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/5/check/openstack-tox-pep8/ac29689/ : SUCCESS in 10m 51s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/5/check/openstack-tox-py27/b5f7b9a/ : SUCCESS in 15m 18s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/5/check/openstack-tox-py35/62908f8/ : SUCCESS in 12m 54s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/5/check/openstack-tox-py36/a02b4e2/ : SUCCESS in 15m 07s\n- openstack-tox-docs http://logs.openstack.org/81/639581/5/check/openstack-tox-docs/2fecb5e/html/ : SUCCESS in 5m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5252585/ : SUCCESS in 44m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/5/check/devstack-plugin-ceph-tempest/9c369a8/ : SUCCESS in 1h 28m 03s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/5/check/neutron-grenade-multinode/4276614/ : SUCCESS in 1h 11m 21s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/5/check/neutron-tempest-linuxbridge/38e52c0/ : SUCCESS in 1h 40m 26s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/5/check/nova-grenade-live-migration/b176521/ : SUCCESS in 1h 14m 17s\n- nova-live-migration http://logs.openstack.org/81/639581/5/check/nova-live-migration/76e6970/ : SUCCESS in 49m 23s\n- nova-next http://logs.openstack.org/81/639581/5/check/nova-next/2612bcc/ : SUCCESS in 2h 00m 31s\n- nova-tox-functional http://logs.openstack.org/81/639581/5/check/nova-tox-functional/d229276/ : SUCCESS in 18m 50s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/5/check/nova-tox-functional-py35/89d78ce/ : SUCCESS in 17m 10s\n- tempest-multinode-full http://logs.openstack.org/81/639581/5/check/tempest-multinode-full/9800559/ : FAILURE in 1h 21m 03s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/5/check/tempest-slow-py3/033a49b/ : SUCCESS in 2h 04m 04s","accounts_in_message":[],"_revision_number":5},{"id":"e017a701be0c2ce25ff1388b0d025be4fe570b2c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-08 07:38:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/5/check/EMC_VxFlexOS_NOVA/f5b9f2f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"472599b8736c8d38aa2795cadacf6431fc3392e0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-08 08:54:30.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28943 : SUCCESS in 1h 47m 10s","accounts_in_message":[],"_revision_number":5},{"id":"74efd8652797652e9d56aa8890f30b6bff206705","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-08 15:26:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/5 : SUCCESS in 2h 31m 39s","accounts_in_message":[],"_revision_number":5},{"id":"8eec8e36773461818259b592d5a120c6d67eec16","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-08 21:48:12.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"f76fc9463a9870aec3a76d3c6ef7a20cf11cabd8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-09 03:39:15.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/5/check-vote/ext-nova-zuul/65e0000 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":5},{"id":"72947bf439f5a8875c74391d38cd6d4538b7f065","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-03-09 04:45:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/81/639581/5/check-nova/Nova-ML2-Sriov/c83ca19 : SUCCESS in 1h 20m 25s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/81/639581/5/check-nova/Nova-MACVTAP-ML2-Sriov/75e960b : SUCCESS in 55m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":5},{"id":"0773914fdd1475fe34c7f18343011bd07bf45b20","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-09 08:52:32.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c02e4d2 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":5},{"id":"888c7926fc07e462ecb406a86a2a2dd6d0d920fd","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2019-03-10 13:12:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"76b32ed0325f2a696a9174ca41b204b0e0dbceda","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2019-03-11 12:24:46.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2bf4652aa20613fa3036e0b4441eff63bea987de","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-18 05:44:49.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f4c0d650941b42805ae777a00d7d4f2d808517ee","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 06:03:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"191c3ea420ca3ec24ab56e1b352020bb0f6581f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 08:08:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/6/check/grenade-py3/87be72d/ : SUCCESS in 1h 01m 09s\n- tempest-full-py3 http://logs.openstack.org/81/639581/6/check/tempest-full-py3/724852d/ : SUCCESS in 1h 31m 09s\n- openstack-tox-cover http://logs.openstack.org/81/639581/6/check/openstack-tox-cover/7da780c/cover/ : SUCCESS in 17m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/6/check/openstack-tox-lower-constraints/dbf0e50/ : SUCCESS in 14m 59s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/6/check/openstack-tox-pep8/38f11cd/ : SUCCESS in 10m 47s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/6/check/openstack-tox-py27/9c4cc2f/ : SUCCESS in 13m 53s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/6/check/openstack-tox-py35/56c197b/ : SUCCESS in 13m 10s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/6/check/openstack-tox-py36/c8f2b41/ : SUCCESS in 12m 03s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/6/check/openstack-tox-py37/d557ac8/ : SUCCESS in 14m 30s\n- openstack-tox-docs http://logs.openstack.org/81/639581/6/check/openstack-tox-docs/2c76a34/html/ : SUCCESS in 6m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/08b8b6e/ : SUCCESS in 50m 22s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/6/check/devstack-plugin-ceph-tempest/a941803/ : SUCCESS in 1h 23m 00s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/6/check/neutron-grenade-multinode/4a5f186/ : SUCCESS in 1h 12m 29s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/6/check/neutron-tempest-linuxbridge/3a85fc8/ : SUCCESS in 1h 37m 53s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/6/check/nova-grenade-live-migration/c588278/ : SUCCESS in 1h 09m 01s\n- nova-live-migration http://logs.openstack.org/81/639581/6/check/nova-live-migration/8d815d0/ : SUCCESS in 49m 46s\n- nova-next http://logs.openstack.org/81/639581/6/check/nova-next/2cb7586/ : SUCCESS in 1h 19m 35s\n- nova-tox-functional http://logs.openstack.org/81/639581/6/check/nova-tox-functional/704b71a/ : SUCCESS in 20m 13s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/6/check/nova-tox-functional-py35/2fd9c2f/ : SUCCESS in 18m 39s\n- tempest-multinode-full http://logs.openstack.org/81/639581/6/check/tempest-multinode-full/7274b88/ : SUCCESS in 1h 48m 46s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/6/check/tempest-slow-py3/1a5d5d5/ : SUCCESS in 2h 13m 37s","accounts_in_message":[],"_revision_number":6},{"id":"97421f4f8963415752412258986393fcde354f60","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-18 09:09:57.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29290 : SUCCESS in 2h 27m 50s","accounts_in_message":[],"_revision_number":6},{"id":"8e45762cc6cb1049de345bcea3346b104ebcf386","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 09:18:54.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/6/check-vote/ext-nova-zuul/8dd14d6 : FAILURE in 55m 16s","accounts_in_message":[],"_revision_number":6},{"id":"7d85ebebe7b687e024d12a109ff6cb195c15e1e1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-18 10:29:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/6 : SUCCESS in 2h 22m 45s","accounts_in_message":[],"_revision_number":6},{"id":"f1d19288dcb7c9f82fa63e4f5ead1e431a30be24","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-18 12:32:28.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"b70191bacb9dade0b6cfd6a9e466e276fd3f75e5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 15:38:29.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/6/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/554d445 : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":6},{"id":"f699039cbe9543ea78dbf890b16c3f71f795c54c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-18 19:31:11.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/6/check/tempest-dsvm-full-xenial/7bb62cf/ : SUCCESS in 1h 13m 05s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/6/check/tempest-dsvm-full-xenial-py3/f289686/ : SUCCESS in 1h 20m 00s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/6/check/grenade-dsvm-xenial/25788cd/ : SUCCESS in 1h 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3dbd35eed5720ba82cfc2f75ec94647b48fd571","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-21 09:37:18.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7d3ead075693c8301eec33a71d6b574fad3373b3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:01:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f4eb6967093698e367852ca6631ac29a02248a0c","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:50:58.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/7/check/nova-out-of-tree-pvm/5f47111 : FAILURE in 38m 09s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/7/check/nova-in-tree-pvm/51cf5c2 : FAILURE in 39m 01s","accounts_in_message":[],"_revision_number":7},{"id":"41b5133806b58bfe15408db87c67875171f925b4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-21 11:34:11.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/7/check/tempest-dsvm-full-xenial/82c4111/ : SUCCESS in 1h 12m 53s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/7/check/tempest-dsvm-full-xenial-py3/f461b93/ : SUCCESS in 1h 14m 47s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/7/check/grenade-dsvm-xenial/cf52ebb/ : SUCCESS in 1h 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dee038ed5168d81abb5087d989481bac632cbf71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 11:53:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/7/check/grenade-py3/1ee21b2/ : SUCCESS in 1h 02m 08s\n- tempest-full-py3 http://logs.openstack.org/81/639581/7/check/tempest-full-py3/bc9b335/ : SUCCESS in 1h 24m 51s\n- openstack-tox-cover http://logs.openstack.org/81/639581/7/check/openstack-tox-cover/eddca58/cover/ : SUCCESS in 15m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/7/check/openstack-tox-lower-constraints/660a2b3/ : SUCCESS in 11m 35s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/7/check/openstack-tox-pep8/fb3421f/ : SUCCESS in 10m 04s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/7/check/openstack-tox-py27/69c4411/ : SUCCESS in 13m 12s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/7/check/openstack-tox-py35/1dc4b55/ : SUCCESS in 13m 30s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/7/check/openstack-tox-py36/297dcc4/ : SUCCESS in 12m 30s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/7/check/openstack-tox-py37/f44ed87/ : SUCCESS in 12m 10s\n- openstack-tox-docs http://logs.openstack.org/81/639581/7/check/openstack-tox-docs/94c7009/html/ : SUCCESS in 7m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9e0c24b/ : SUCCESS in 47m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/7/check/devstack-plugin-ceph-tempest/167ca39/ : SUCCESS in 1h 28m 37s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/7/check/neutron-grenade-multinode/7653122/ : SUCCESS in 1h 26m 51s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/7/check/neutron-tempest-linuxbridge/0a0b0f5/ : SUCCESS in 1h 48m 12s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/7/check/nova-grenade-live-migration/1b3b90f/ : SUCCESS in 1h 09m 57s\n- nova-live-migration http://logs.openstack.org/81/639581/7/check/nova-live-migration/e24717b/ : SUCCESS in 51m 37s\n- nova-next http://logs.openstack.org/81/639581/7/check/nova-next/79dfaae/ : SUCCESS in 1h 15m 32s\n- nova-tox-functional http://logs.openstack.org/81/639581/7/check/nova-tox-functional/3320683/ : SUCCESS in 20m 26s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/7/check/nova-tox-functional-py35/4447c61/ : SUCCESS in 19m 41s\n- tempest-multinode-full http://logs.openstack.org/81/639581/7/check/tempest-multinode-full/6ca5a6c/ : SUCCESS in 1h 44m 28s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/81/639581/7/check/tempest-slow-py3/809fbb4/ : SUCCESS in 2h 04m 51s","accounts_in_message":[],"_revision_number":7},{"id":"a07c74e2fcbc54f9a264cc14b513c88b0fef5c28","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 11:55:06.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/7/check/EMC_VxFlexOS_NOVA/dc33d38/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"7c7bea75d5622bd68f4eb0e4ca65755c05e7a819","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-21 13:37:28.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29433 : FAILURE in 3h 26m 08s","accounts_in_message":[],"_revision_number":7},{"id":"5ea6a5aa69e49ee16570e7c623c29c1caa973850","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-21 14:24:36.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/7 : FAILURE in 2h 22m 51s","accounts_in_message":[],"_revision_number":7},{"id":"d18a4f946a5997a860ca0eeb390d63fcd585198c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-21 17:16:29.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"24c2b08a1954290be53e95a379925115f1d87c93","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-21 17:47:35.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"488c3fdf82dab68f2822633e27087a6b6b3a5f9b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 20:02:25.000000000","message":"Patch Set 7:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/7/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9e343ae : SUCCESS in 1h 32m 36s","accounts_in_message":[],"_revision_number":7},{"id":"7c32d118ed799683300c47dc76d716293ad4ebfe","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-26 20:11:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"787fe375c919abc346f88e9e1ee3c61b84e894a3","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-27 04:38:51.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d28a8770573f8ac10ef862996f2250ad6515e435","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 04:56:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4516b3ef154bf7de371f2961e1aaf96a2412c115","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-27 06:58:50.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/8/check/tempest-dsvm-full-xenial/a130dbe/ : SUCCESS in 1h 51m 34s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/8/check/tempest-dsvm-full-xenial-py3/61409c8/ : SUCCESS in 1h 13m 04s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/8/check/grenade-dsvm-xenial/ecd9799/ : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"57652489d60dc4def8b9b91ff9e0480058514a49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 07:10:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/8/check/grenade-py3/b374c8b/ : SUCCESS in 53m 28s\n- tempest-full-py3 http://logs.openstack.org/81/639581/8/check/tempest-full-py3/d2e56a5/ : SUCCESS in 1h 14m 58s\n- openstack-tox-cover http://logs.openstack.org/81/639581/8/check/openstack-tox-cover/12376ee/cover/ : SUCCESS in 15m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/8/check/openstack-tox-lower-constraints/02d5fc5/ : SUCCESS in 12m 28s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/8/check/openstack-tox-pep8/284dff7/ : SUCCESS in 11m 03s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/8/check/openstack-tox-py27/145a3be/ : SUCCESS in 12m 52s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/8/check/openstack-tox-py35/e29d085/ : SUCCESS in 14m 01s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/8/check/openstack-tox-py36/19d9302/ : SUCCESS in 12m 34s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/8/check/openstack-tox-py37/6446e7b/ : SUCCESS in 13m 06s\n- openstack-tox-docs http://logs.openstack.org/81/639581/8/check/openstack-tox-docs/7cc1261/html/ : SUCCESS in 6m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/8/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/76e0705/ : SUCCESS in 48m 32s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/8/check/devstack-plugin-ceph-tempest/18a49b9/ : SUCCESS in 1h 26m 40s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/8/check/neutron-grenade-multinode/c2adb09/ : SUCCESS in 1h 08m 28s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/8/check/neutron-tempest-linuxbridge/35907bf/ : SUCCESS in 1h 47m 36s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/8/check/nova-grenade-live-migration/6c153f5/ : SUCCESS in 1h 03m 17s\n- nova-live-migration http://logs.openstack.org/81/639581/8/check/nova-live-migration/b640dd2/ : SUCCESS in 46m 27s\n- nova-next http://logs.openstack.org/81/639581/8/check/nova-next/65c25b6/ : SUCCESS in 1h 13m 09s\n- nova-tox-functional http://logs.openstack.org/81/639581/8/check/nova-tox-functional/4b83b0f/ : SUCCESS in 19m 14s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/8/check/nova-tox-functional-py35/ec4b96f/ : SUCCESS in 20m 37s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/8/check/tempest-slow-py3/1438398/ : SUCCESS in 2h 13m 50s","accounts_in_message":[],"_revision_number":8},{"id":"539fcabe71e0c19ff37026342194a0151ac32baa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:10:55.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"aa68d39e20200bc81c6cb85b2b8180557e59325c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 07:12:35.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/8/check/EMC_VxFlexOS_NOVA/3303614/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"f7ea48715d07f6f4c1f3f8ea9729100bbbd191b1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-27 07:41:43.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29585 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":8},{"id":"8e4bf7f0e073c6dc92ef9ad2aed344125d5aa2ba","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-27 08:27:23.000000000","message":"Patch Set 8:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/8/check/nova-out-of-tree-pvm/9693911 : FAILURE in 2h 14m 04s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/8/check/nova-in-tree-pvm/1463f17 : FAILURE in 3h 25m 31s","accounts_in_message":[],"_revision_number":8},{"id":"97a84cd0e9b67c746c841167a5d77f595fff689b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-27 09:04:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/8 : SUCCESS in 2h 14m 08s","accounts_in_message":[],"_revision_number":8},{"id":"6992b3e2662391fbb55f4bd894887f47d6333bf3","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2019-03-27 09:47:25.000000000","message":"Patch Set 8:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":8},{"id":"4c4359b2ad525b7b75014b0f4c6e3f5271f8eb21","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 12:02:41.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/8/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/b90a6ae : FAILURE in 22m 46s","accounts_in_message":[],"_revision_number":8},{"id":"a850df47fe4facf074c6dc3063dd24e56b5d5522","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-27 12:23:51.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"4388bf34488c864f148c82d73e1db43458fe6ac8","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 04:13:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"0fbaf87b3209dd1db49cdb61add2b0f685f8720b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 04:43:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"25f3ef144bd8f3ecbfe272c038252a7ac55c0501","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-29 06:09:26.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"41340754cc71a1be49afff073ab60ed052b7e132","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 06:45:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"10cc2d015c737c20db31ea8c250ec9cea5177674","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 08:36:57.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/10/check/tempest-dsvm-full-xenial/18238cb/ : SUCCESS in 1h 08m 14s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/10/check/tempest-dsvm-full-xenial-py3/eb18ff2/ : SUCCESS in 1h 16m 48s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/10/check/grenade-dsvm-xenial/7bdc8f0/ : SUCCESS in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"68618f4936ce0b63c2b74bc17dee9fbfec87d1c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 09:02:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/10/check/grenade-py3/c67e00f/ : SUCCESS in 55m 51s\n- tempest-full-py3 http://logs.openstack.org/81/639581/10/check/tempest-full-py3/ad1ff44/ : SUCCESS in 1h 20m 10s\n- openstack-tox-cover http://logs.openstack.org/81/639581/10/check/openstack-tox-cover/cddd369/cover/ : SUCCESS in 15m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/10/check/openstack-tox-lower-constraints/9a26792/ : SUCCESS in 11m 51s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/10/check/openstack-tox-pep8/2d46a73/ : FAILURE in 8m 47s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/10/check/openstack-tox-py27/ff55c59/ : SUCCESS in 14m 17s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/10/check/openstack-tox-py35/53098fc/ : SUCCESS in 13m 08s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/10/check/openstack-tox-py36/1a8b203/ : SUCCESS in 15m 00s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/10/check/openstack-tox-py37/3080301/ : SUCCESS in 14m 48s\n- openstack-tox-docs http://logs.openstack.org/81/639581/10/check/openstack-tox-docs/44a223a/html/ : SUCCESS in 7m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/10/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c7913a4/ : SUCCESS in 1h 01m 16s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/10/check/devstack-plugin-ceph-tempest/0a43992/ : SUCCESS in 1h 45m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/10/check/neutron-grenade-multinode/481a03c/ : SUCCESS in 1h 10m 50s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/10/check/neutron-tempest-linuxbridge/442b89c/ : SUCCESS in 1h 31m 00s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/10/check/nova-grenade-live-migration/97afa4e/ : SUCCESS in 1h 07m 17s\n- nova-live-migration http://logs.openstack.org/81/639581/10/check/nova-live-migration/18f469d/ : SUCCESS in 48m 48s\n- nova-next http://logs.openstack.org/81/639581/10/check/nova-next/7774ee3/ : SUCCESS in 1h 26m 30s\n- nova-tox-functional http://logs.openstack.org/81/639581/10/check/nova-tox-functional/8d3997d/ : SUCCESS in 17m 04s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/10/check/nova-tox-functional-py35/bc9751c/ : SUCCESS in 20m 06s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/10/check/tempest-slow-py3/213384b/ : SUCCESS in 2h 19m 07s","accounts_in_message":[],"_revision_number":10},{"id":"fd250aefb5807ecfedf9c7305256f6611b1c7899","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 10:54:43.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29682 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":10},{"id":"2ecdd9184ab2b802045e212f9d08fc8bdd06fbd6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 11:02:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/10 : SUCCESS in 2h 12m 17s","accounts_in_message":[],"_revision_number":10},{"id":"949772d806588c092a7774daa1f33a93e2a5515d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-29 15:03:24.000000000","message":"Patch Set 10:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/10/check/nova-out-of-tree-pvm/13d7bc5 : FAILURE in 2h 46m 03s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/10/check/nova-in-tree-pvm/145b215 : FAILURE in 3h 24m 28s","accounts_in_message":[],"_revision_number":10},{"id":"5410f30493770aaf6063518bc0161d73bd722be2","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 15:23:51.000000000","message":"Patch Set 10:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/06c3e61 : SUCCESS in 1h 48m 30s","accounts_in_message":[],"_revision_number":10},{"id":"cf586312657a7182d1fbcbfe4e06cc7cb8d3570e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 20:45:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"738e243e247281b91f8180244d55e67b49b60be9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 21:16:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e4a0c03860e19634d7c211be0f57d98594a45205","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 23:16:21.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/11/check/tempest-dsvm-full-xenial/4112146/ : SUCCESS in 1h 27m 37s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/11/check/tempest-dsvm-full-xenial-py3/00a0b24/ : SUCCESS in 1h 19m 06s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/11/check/grenade-dsvm-xenial/dd2141c/ : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"489ccbecda00455c20f81c6f908a45abac9ad795","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 23:30:36.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/11 : FAILURE in 17m 42s","accounts_in_message":[],"_revision_number":11},{"id":"6546d22d2786090dc94de9bf2c6130cd43d358dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 23:53:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/11/check/grenade-py3/cd57377/ : SUCCESS in 51m 20s\n- tempest-full-py3 http://logs.openstack.org/81/639581/11/check/tempest-full-py3/be4b672/ : SUCCESS in 1h 36m 33s\n- openstack-tox-cover http://logs.openstack.org/81/639581/11/check/openstack-tox-cover/6baed04/cover/ : SUCCESS in 15m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/11/check/openstack-tox-lower-constraints/f5d2b3f/ : SUCCESS in 12m 39s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/11/check/openstack-tox-pep8/fd31659/ : POST_FAILURE in 15m 59s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/11/check/openstack-tox-py27/87b6331/ : SUCCESS in 13m 38s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/11/check/openstack-tox-py35/e66de16/ : SUCCESS in 13m 22s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/11/check/openstack-tox-py36/96fca3a/ : SUCCESS in 11m 06s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/11/check/openstack-tox-py37/aa7504b/ : SUCCESS in 16m 24s\n- openstack-tox-docs http://logs.openstack.org/81/639581/11/check/openstack-tox-docs/33348ad/html/ : SUCCESS in 7m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/11/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/4449c34/ : SUCCESS in 52m 31s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/11/check/devstack-plugin-ceph-tempest/0965931/ : SUCCESS in 1h 23m 42s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/11/check/neutron-grenade-multinode/600e1be/ : SUCCESS in 1h 09m 40s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/11/check/neutron-tempest-linuxbridge/935dd69/ : POST_FAILURE in 1h 27m 48s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/11/check/nova-grenade-live-migration/426eae7/ : SUCCESS in 57m 49s\n- nova-live-migration http://logs.openstack.org/81/639581/11/check/nova-live-migration/6ff8dd8/ : SUCCESS in 43m 35s\n- nova-next http://logs.openstack.org/81/639581/11/check/nova-next/95aa491/ : SUCCESS in 1h 36m 39s\n- nova-tox-functional http://logs.openstack.org/81/639581/11/check/nova-tox-functional/f57edb1/ : SUCCESS in 22m 08s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/11/check/nova-tox-functional-py35/859d064/ : SUCCESS in 22m 50s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/11/check/tempest-slow-py3/11a2336/ : SUCCESS in 2h 29m 04s","accounts_in_message":[],"_revision_number":11},{"id":"07f2001d52d77d6e38fd14ab90c9a7f0b194a5a1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-30 00:48:18.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29724 : SUCCESS in 2h 15m 17s","accounts_in_message":[],"_revision_number":11},{"id":"8b596ebf865f993864b86bb5a4179308ac94e5da","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-30 01:05:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/11/check/nova-out-of-tree-pvm/577d0ee : SUCCESS in 2h 29m 28s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/11/check/nova-in-tree-pvm/8bb515d : SUCCESS in 2h 36m 37s","accounts_in_message":[],"_revision_number":11},{"id":"9f3fcb28d438371bec75ea9759af43a305ec6ab9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 18:22:22.000000000","message":"Patch Set 11:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/11/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/a965773 : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":11},{"id":"05567b11deb76d6475a0cb87cc8a54fe4293e957","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-01 21:18:06.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"ff9173c93b7299e7b143aed1e5044ac729d24c89","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 21:56:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ef60ed38ef9c9dc1e1c5d1d57203b3cab23d0168","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-01 23:56:20.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/12/check/tempest-dsvm-full-xenial/c0dafef/ : SUCCESS in 1h 13m 57s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/12/check/tempest-dsvm-full-xenial-py3/2ec8335/ : FAILURE in 19m 09s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/12/check/grenade-dsvm-xenial/130057a/ : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"40d4847b47e873b71ffc7761a3fe07a94f2f61fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 00:47:03.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/12/check/grenade-py3/c5496b0/ : SUCCESS in 56m 41s\n- tempest-full-py3 http://logs.openstack.org/81/639581/12/check/tempest-full-py3/d79abb1/ : SUCCESS in 1h 24m 21s\n- openstack-tox-cover http://logs.openstack.org/81/639581/12/check/openstack-tox-cover/5e67c8d/ : FAILURE in 15m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/12/check/openstack-tox-lower-constraints/b30f385/ : FAILURE in 13m 22s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/12/check/openstack-tox-pep8/66baf53/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/12/check/openstack-tox-py27/75ce8c9/ : SUCCESS in 14m 19s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/12/check/openstack-tox-py35/cec7710/ : FAILURE in 12m 05s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/12/check/openstack-tox-py36/ebb000a/ : FAILURE in 15m 57s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/12/check/openstack-tox-py37/9abb14d/ : FAILURE in 13m 20s\n- openstack-tox-docs http://logs.openstack.org/81/639581/12/check/openstack-tox-docs/747097a/html/ : SUCCESS in 7m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/12/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6dc9507/ : SUCCESS in 59m 25s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/12/check/devstack-plugin-ceph-tempest/858e887/ : SUCCESS in 1h 32m 39s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/12/check/neutron-grenade-multinode/711838e/ : SUCCESS in 1h 08m 55s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/12/check/neutron-tempest-linuxbridge/1f291bd/ : SUCCESS in 1h 40m 34s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/12/check/nova-grenade-live-migration/4b798d6/ : SUCCESS in 1h 03m 28s\n- nova-live-migration http://logs.openstack.org/81/639581/12/check/nova-live-migration/1e7a69c/ : FAILURE in 44m 05s\n- nova-next http://logs.openstack.org/81/639581/12/check/nova-next/53cfe76/ : SUCCESS in 1h 23m 17s\n- nova-tox-functional http://logs.openstack.org/81/639581/12/check/nova-tox-functional/8ae7ef2/ : SUCCESS in 21m 01s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/12/check/nova-tox-functional-py35/01613ed/ : SUCCESS in 23m 14s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/12/check/tempest-slow-py3/313443d/ : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":12},{"id":"019e453ee29244a0a239b8d62c8eee4fc0c3faa3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 01:17:07.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29799 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":12},{"id":"9d660f861834e8d02672876c49f0e757a296e0f1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 02:57:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/12 : SUCCESS in 2h 14m 40s","accounts_in_message":[],"_revision_number":12},{"id":"7df7fcf9b17c3ba7e84b0feca0490fcad172d506","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 03:23:49.000000000","message":"Patch Set 12:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/12/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/852d843 : FAILURE in 21m 27s","accounts_in_message":[],"_revision_number":12},{"id":"67dcbaa730a9f4d63d6ef378485e134c55f75a7a","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-02 10:13:10.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"d7575bde1da8e2425fa4e4ae5380910f2dbc7d14","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 10:58:44.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"49bd896d2fb20795c11f24056948a727495dd197","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 11:10:10.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/13 : FAILURE in 47s","accounts_in_message":[],"_revision_number":13},{"id":"1c3abe6982a9525b954eadd488a5fcfc9972e92a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 13:41:18.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/13/check/grenade-py3/0abeb3e/ : SUCCESS in 1h 01m 47s\n- tempest-full-py3 http://logs.openstack.org/81/639581/13/check/tempest-full-py3/5d3fa1a/ : SUCCESS in 1h 30m 48s\n- openstack-tox-cover http://logs.openstack.org/81/639581/13/check/openstack-tox-cover/81e608a/ : FAILURE in 17m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/13/check/openstack-tox-lower-constraints/28cc294/ : FAILURE in 12m 48s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/13/check/openstack-tox-pep8/a386046/ : SUCCESS in 11m 21s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/13/check/openstack-tox-py27/233e936/ : SUCCESS in 19m 40s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/13/check/openstack-tox-py35/7c3f5bf/ : FAILURE in 13m 02s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/13/check/openstack-tox-py36/94a357d/ : FAILURE in 16m 48s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/13/check/openstack-tox-py37/f83cf86/ : FAILURE in 13m 43s\n- openstack-tox-docs http://logs.openstack.org/81/639581/13/check/openstack-tox-docs/fb5ccca/html/ : SUCCESS in 9m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/13/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b5689c2/ : SUCCESS in 52m 21s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/13/check/devstack-plugin-ceph-tempest/6644a34/ : SUCCESS in 1h 32m 31s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/13/check/neutron-grenade-multinode/f2ae8a2/ : SUCCESS in 1h 02m 31s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/13/check/neutron-tempest-linuxbridge/06b6105/ : SUCCESS in 1h 21m 47s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/13/check/nova-grenade-live-migration/57ae229/ : SUCCESS in 1h 09m 28s\n- nova-live-migration http://logs.openstack.org/81/639581/13/check/nova-live-migration/fb8b58f/ : SUCCESS in 52m 26s\n- nova-next http://logs.openstack.org/81/639581/13/check/nova-next/fb8fb54/ : SUCCESS in 1h 46m 24s\n- nova-tox-functional http://logs.openstack.org/81/639581/13/check/nova-tox-functional/bb8759e/ : SUCCESS in 19m 36s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/13/check/nova-tox-functional-py35/2b0df04/ : SUCCESS in 20m 28s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/13/check/tempest-slow-py3/213530c/ : SUCCESS in 2h 19m 25s","accounts_in_message":[],"_revision_number":13},{"id":"66cafaf515bbe81c3fa3792c4f502b3b47ff1d57","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 14:28:37.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29888 : FAILURE in 21m 47s","accounts_in_message":[],"_revision_number":13},{"id":"00e69ed818e7a9b176aaa769bcc4342ca9112521","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-02 21:00:05.000000000","message":"Patch Set 13:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/13/check/tempest-dsvm-full-xenial/dbeedf5/ : SUCCESS in 1h 21m 46s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/13/check/tempest-dsvm-full-xenial-py3/0786672/ : SUCCESS in 1h 21m 36s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/13/check/grenade-dsvm-xenial/15543aa/ : SUCCESS in 1h 01m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8da74a3a8c8793ab7cfa1b24e17459de3a0f33dc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-03 13:40:20.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/75a38f0 : SUCCESS in 2h 22m 36s","accounts_in_message":[],"_revision_number":13},{"id":"a27423508308ea1ff5da883752aaf0aeabb13872","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 08:44:44.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"c5f599863dc048fe5c4396e17d720639ef0484b5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-04 09:25:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cabca78d34bedefe77b1cf94437e0e27505e1da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 12:00:23.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/14/check/grenade-py3/9e958f2/ : SUCCESS in 1h 12m 09s\n- tempest-full-py3 http://logs.openstack.org/81/639581/14/check/tempest-full-py3/4043a11/ : SUCCESS in 1h 36m 24s\n- openstack-tox-cover http://logs.openstack.org/81/639581/14/check/openstack-tox-cover/e91f523/ : FAILURE in 20m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/14/check/openstack-tox-lower-constraints/0a28d56/ : FAILURE in 15m 09s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/14/check/openstack-tox-pep8/9fe3111/ : SUCCESS in 10m 22s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/14/check/openstack-tox-py27/351bd7d/ : SUCCESS in 13m 19s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/14/check/openstack-tox-py35/31da392/ : FAILURE in 13m 28s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/14/check/openstack-tox-py36/4e1f74a/ : FAILURE in 12m 21s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/14/check/openstack-tox-py37/48511b0/ : FAILURE in 13m 38s\n- openstack-tox-docs http://logs.openstack.org/81/639581/14/check/openstack-tox-docs/0e515b9/html/ : SUCCESS in 11m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/14/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/9f7b1f0/ : SUCCESS in 1h 07m 09s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/14/check/devstack-plugin-ceph-tempest/6394848/ : SUCCESS in 1h 37m 57s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/14/check/neutron-grenade-multinode/d26992a/ : SUCCESS in 1h 10m 29s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/14/check/neutron-tempest-linuxbridge/ac9d5c9/ : SUCCESS in 1h 38m 24s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/14/check/nova-grenade-live-migration/4dce6be/ : SUCCESS in 1h 18m 35s\n- nova-live-migration http://logs.openstack.org/81/639581/14/check/nova-live-migration/520570a/ : SUCCESS in 47m 34s\n- nova-next http://logs.openstack.org/81/639581/14/check/nova-next/9c5d4b1/ : SUCCESS in 1h 37m 15s\n- nova-tox-functional http://logs.openstack.org/81/639581/14/check/nova-tox-functional/24cce9e/ : SUCCESS in 18m 33s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/14/check/nova-tox-functional-py35/6b903ba/ : SUCCESS in 20m 26s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/14/check/tempest-slow-py3/97aa6e7/ : SUCCESS in 2h 04m 17s","accounts_in_message":[],"_revision_number":14},{"id":"17ec6391e1de60137a44964f018c16afbc6101b0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-04 18:11:16.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30036 : SUCCESS in 3h 16m 55s","accounts_in_message":[],"_revision_number":14},{"id":"71ccc1c0ebd7130ecbaa953ad39d5248d8119c6a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 18:32:46.000000000","message":"Patch Set 14:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/14/check/tempest-dsvm-full-xenial/d2da290/ : SUCCESS in 1h 20m 08s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/14/check/tempest-dsvm-full-xenial-py3/0c9559b/ : SUCCESS in 1h 18m 45s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/14/check/grenade-dsvm-xenial/b435b3f/ : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f7b5c5ae2ce2b5134b8f253e69fbe27756da1384","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 20:44:48.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"528134beba4d40af6bd4dcd34a6d15a34fb2b796","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-04 23:30:16.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/15/check/tempest-dsvm-full-xenial/cb925db/ : SUCCESS in 1h 04m 09s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/15/check/tempest-dsvm-full-xenial-py3/01bd3cc/ : SUCCESS in 1h 16m 17s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/15/check/grenade-dsvm-xenial/e7d7cbe/ : SUCCESS in 57m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"067483d33c194de9f5fc5fa41b19eb328031ca87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 00:41:17.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/15/check/grenade-py3/ddab8d4/ : FAILURE in 43m 02s\n- tempest-full-py3 http://logs.openstack.org/81/639581/15/check/tempest-full-py3/6b61a2f/ : SUCCESS in 1h 40m 26s\n- openstack-tox-cover http://logs.openstack.org/81/639581/15/check/openstack-tox-cover/6e6b118/cover/ : SUCCESS in 17m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/15/check/openstack-tox-lower-constraints/357ad89/ : SUCCESS in 22m 33s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/15/check/openstack-tox-pep8/e484b92/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/15/check/openstack-tox-py27/38d7f02/ : SUCCESS in 12m 28s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/15/check/openstack-tox-py35/20a6d44/ : SUCCESS in 13m 51s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/15/check/openstack-tox-py36/5124c9c/ : SUCCESS in 13m 13s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/15/check/openstack-tox-py37/a769bd3/ : SUCCESS in 15m 48s\n- openstack-tox-docs http://logs.openstack.org/81/639581/15/check/openstack-tox-docs/a957124/html/ : SUCCESS in 7m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/15/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/43b7796/ : SUCCESS in 49m 38s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/15/check/devstack-plugin-ceph-tempest/e7ae602/ : TIMED_OUT in 2h 05m 39s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/15/check/neutron-grenade-multinode/2c13612/ : SUCCESS in 1h 13m 59s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/15/check/neutron-tempest-linuxbridge/24e9bac/ : SUCCESS in 1h 24m 51s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/15/check/nova-grenade-live-migration/93bd803/ : SUCCESS in 1h 04m 21s\n- nova-live-migration http://logs.openstack.org/81/639581/15/check/nova-live-migration/b0722e2/ : SUCCESS in 58m 16s\n- nova-next http://logs.openstack.org/81/639581/15/check/nova-next/977d38e/ : SUCCESS in 1h 32m 19s\n- nova-tox-functional http://logs.openstack.org/81/639581/15/check/nova-tox-functional/e8e6355/ : SUCCESS in 20m 37s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/15/check/nova-tox-functional-py35/154fbf3/ : SUCCESS in 21m 43s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/15/check/tempest-slow-py3/aac7e32/ : SUCCESS in 2h 52m 06s","accounts_in_message":[],"_revision_number":15},{"id":"406f4da49743d15d6fd665964464394cda7ead3d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-05 00:52:10.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30077 : FAILURE in 1h 45m 47s","accounts_in_message":[],"_revision_number":15},{"id":"cd161d0579564ba343e584f6b394386d4ec86d73","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-05 02:36:57.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/15 : SUCCESS in 2h 14m 35s","accounts_in_message":[],"_revision_number":15},{"id":"b5eb4f54377dd0a6d8f092cb6714caccc403d01a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-05 16:46:06.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/15/check-vote/ext-nova-zuul/c97e6bc : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":15},{"id":"cef23fed6ec3ca363c84d4aa89da3c81a19706d5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-07 07:22:11.000000000","message":"Patch Set 15:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/15/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/c0c8ce7 : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":15},{"id":"abecf1de0ca8c473d647e3c19cfc0d9d90523c21","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-08 04:20:02.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/15/check/check-dsvm-tempest-vz7-exe-minimal/98ff618 : FAILURE in 1h 01m 40s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":15},{"id":"dea849d45ae31da35f0c56f9b312e79519be31bf","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-12 06:57:52.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"4d754f5c8e5acd69b090a6335a090bcfc6edb3b0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 07:25:33.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"475b902fc0087015bfc9bc79556acbe93abb472f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 09:35:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/16/check/grenade-py3/ce21187/ : SUCCESS in 1h 01m 56s\n- tempest-full-py3 http://logs.openstack.org/81/639581/16/check/tempest-full-py3/d9561de/ : SUCCESS in 1h 31m 59s\n- openstack-tox-cover http://logs.openstack.org/81/639581/16/check/openstack-tox-cover/e707814/cover/ : SUCCESS in 16m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/16/check/openstack-tox-lower-constraints/cfa6dfb/ : SUCCESS in 15m 39s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/16/check/openstack-tox-pep8/225fbba/ : SUCCESS in 10m 10s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/16/check/openstack-tox-py27/ed9eebc/ : SUCCESS in 14m 25s\n- openstack-tox-py35 http://logs.openstack.org/81/639581/16/check/openstack-tox-py35/c7f9197/ : SUCCESS in 14m 43s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/16/check/openstack-tox-py36/d6cb8ca/ : SUCCESS in 14m 25s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/16/check/openstack-tox-py37/eb3bbfc/ : SUCCESS in 15m 11s\n- openstack-tox-docs http://logs.openstack.org/81/639581/16/check/openstack-tox-docs/b198aa2/html/ : SUCCESS in 6m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/16/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/df1e30b/ : SUCCESS in 55m 12s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/16/check/devstack-plugin-ceph-tempest/53f0126/ : SUCCESS in 1h 28m 54s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/16/check/neutron-grenade-multinode/548f4e3/ : SUCCESS in 1h 17m 42s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/16/check/neutron-tempest-linuxbridge/3ab3a5a/ : SUCCESS in 1h 44m 02s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/16/check/nova-grenade-live-migration/965c3a7/ : SUCCESS in 1h 14m 19s\n- nova-live-migration http://logs.openstack.org/81/639581/16/check/nova-live-migration/2b9f54b/ : SUCCESS in 57m 12s\n- nova-next http://logs.openstack.org/81/639581/16/check/nova-next/0bad189/ : SUCCESS in 1h 29m 52s\n- nova-tox-functional http://logs.openstack.org/81/639581/16/check/nova-tox-functional/adb5722/ : SUCCESS in 20m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/81/639581/16/check/nova-tox-functional-py35/affb948/ : SUCCESS in 20m 43s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/16/check/tempest-slow-py3/a80d79e/ : SUCCESS in 2h 17m 30s","accounts_in_message":[],"_revision_number":16},{"id":"3e89acad13628cbfe45283256326d71398912520","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 09:37:07.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/16/check/EMC_VxFlexOS_NOVA/4e13c89/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":16},{"id":"d140126fd65c6ebd5ad8718a56b0cb30bf657016","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-12 10:37:17.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/16 : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":16},{"id":"d229f917d0cf84a42728c67b9185648612b1b7b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 13:33:43.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":16},{"id":"7a8a61ca0c318f6a441a22e5d3e0963c3308405d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-12 17:30:32.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30435 : SUCCESS in 2h 19m 34s","accounts_in_message":[],"_revision_number":16},{"id":"a32cfb333ccb8414415e53437a77e2999b4bc8b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-12 21:32:49.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/16/check/tempest-dsvm-full-xenial/8f65487/ : SUCCESS in 1h 26m 58s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/16/check/tempest-dsvm-full-xenial-py3/14aaaaf/ : SUCCESS in 1h 20m 02s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/16/check/grenade-dsvm-xenial/14e62bc/ : SUCCESS in 1h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"31799760a96de3a2e94217666738ab43b046b276","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-13 07:06:37.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"664360d8a79a2c8a94c6782e54face222f06c0fb","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-13 11:44:25.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"4adc15a4df0be2840ce0b8cbfe2219891a88ae9c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-16 20:30:26.000000000","message":"Patch Set 16:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/16/check/check-dsvm-tempest-vz7-exe-minimal/a48be10 : FAILURE in 19m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":16},{"id":"9b8a8ce2031d374cb48611225cb78488257b7cdf","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-24 01:44:56.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"0fa4937dc276bced22099378a31a036453a72241","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:07:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0e64e5366fd3cb9e6850f13364c082ae3e59cd1e","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:14:11.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/17/check/check-dsvm-tempest-vz7-exe-minimal/fe1e9b1 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"1a021ae08a2f3501a17f5600c4edaa638a448a74","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-24 02:20:08.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30975 : FAILURE in 8s","accounts_in_message":[],"_revision_number":17},{"id":"52594c97ce8091249e63353c753cce295b71f6e5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 02:32:54.000000000","message":"Patch Set 17:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/17/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/75ad49b : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":17},{"id":"5eec489ccd127e3311086c2bde77f5502465cf6e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-24 02:36:52.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/17/check/tempest-dsvm-full-xenial/c1838cd/ : FAILURE in 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/17/check/tempest-dsvm-full-xenial-py3/73dd566/ : FAILURE in 9s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/17/check/grenade-dsvm-xenial/3b29893/ : FAILURE in 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1beb607bb2c36cbbc9602c5c5ad6e29b27666e02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 06:57:50.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/17/check/grenade-py3/a83335e/ : SUCCESS in 51m 46s\n- tempest-full-py3 http://logs.openstack.org/81/639581/17/check/tempest-full-py3/d5b8565/ : SUCCESS in 1h 40m 48s\n- openstack-tox-cover http://logs.openstack.org/81/639581/17/check/openstack-tox-cover/b4e7762/cover/ : SUCCESS in 17m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/17/check/openstack-tox-lower-constraints/b901e20/ : SUCCESS in 13m 21s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/17/check/openstack-tox-pep8/8cbf29d/ : SUCCESS in 9m 56s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/17/check/openstack-tox-py27/51e5dfc/ : SUCCESS in 11m 32s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/17/check/openstack-tox-py36/ef00d58/ : SUCCESS in 11m 51s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/17/check/openstack-tox-py37/775380f/ : SUCCESS in 15m 33s\n- openstack-tox-docs http://logs.openstack.org/81/639581/17/check/openstack-tox-docs/d7f490f/html/ : SUCCESS in 6m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/17/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c5d80cb/ : SUCCESS in 51m 14s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/17/check/devstack-plugin-ceph-tempest/02782c0/ : SUCCESS in 1h 42m 47s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/17/check/neutron-grenade-multinode/8e9a2e2/ : SUCCESS in 1h 05m 38s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/17/check/neutron-tempest-linuxbridge/47f4f92/ : SUCCESS in 1h 52m 42s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/17/check/nova-grenade-live-migration/9310d9d/ : SUCCESS in 1h 00m 14s\n- nova-live-migration http://logs.openstack.org/81/639581/17/check/nova-live-migration/88836b8/ : SUCCESS in 43m 35s\n- nova-next http://logs.openstack.org/81/639581/17/check/nova-next/d80f0f3/ : SUCCESS in 1h 21m 40s\n- nova-tox-functional http://logs.openstack.org/81/639581/17/check/nova-tox-functional/5a09c11/ : SUCCESS in 23m 05s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639581/17/check/nova-tox-functional-py36/050d3cd/ : SUCCESS in 21m 47s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/17/check/tempest-slow-py3/5dbd0ad/ : SUCCESS in 2h 09m 09s","accounts_in_message":[],"_revision_number":17},{"id":"793e8f47b2d813d408e1ab4ea2c155c004a6f85b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-24 06:59:58.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/17/check/EMC_VxFlexOS_NOVA/ecf7e87/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":17},{"id":"589e8a0b06e0032330f40f4aada060197e1b2d3d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-24 07:07:56.000000000","message":"Patch Set 17:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/17/check/nova-out-of-tree-pvm/a6fc985 : SUCCESS in 1h 50m 21s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/17/check/nova-in-tree-pvm/4136a72 : FAILURE in 1h 37m 50s","accounts_in_message":[],"_revision_number":17},{"id":"7cebf25ee030759e166a310600b6719ada4d8aaf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-24 07:30:26.000000000","message":"Patch Set 17:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-17 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":17},{"id":"b7ea57cf13c5d13cffc8f52cc0c01e449dc86e2c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-24 07:31:50.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/17 : FAILURE in 2h 20m 20s","accounts_in_message":[],"_revision_number":17},{"id":"52b0daf9ad79a08448465e6a3d82c757b81a116a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-24 07:55:09.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/17/check-vote/ext-nova-zuul/7e8bee3 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":17},{"id":"f5a7ca77c384005f9a1b157eeffffcc06fc8fcb9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:30:51.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"dd55c3812b4d47005566628d5f7dde2c7895f707","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:51:36.000000000","message":"Uploaded patch set 20: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"cdab4448c19f081a481fe22b55a088d7755c08ad","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-25 20:45:57.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/20/check/tempest-dsvm-full-xenial/6d87a87/ : SUCCESS in 2h 05m 10s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/20/check/tempest-dsvm-full-xenial-py3/4c22c2f/ : SUCCESS in 1h 27m 47s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/20/check/grenade-dsvm-xenial/1d34757/ : SUCCESS in 1h 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d19906882574b6c11e19e8d4f0bdc5a0ba9caa1f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-25 22:59:34.000000000","message":"Patch Set 20:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/20/check-vote/ext-nova-zuul/5f5e26a : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":20},{"id":"0cc11fa9466a9e76baeddc5fb46df0c3dbd5169e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 00:45:41.000000000","message":"Patch Set 20:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/20/check/nova-out-of-tree-pvm/c64e2bc : FAILURE in 3h 48m 02s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/20/check/nova-in-tree-pvm/d68ccf7 : FAILURE in 5h 00m 52s","accounts_in_message":[],"_revision_number":20},{"id":"a7f0cdaf11e2ed6f5eaf91a2f91336729896d0c8","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 01:46:02.000000000","message":"Patch Set 20:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/20/check/check-dsvm-tempest-vz7-exe-minimal/45a0b26 : FAILURE in 12m 26s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":20},{"id":"7b6297e59b1d60d34f0d55c84e68029a6452886c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 02:37:03.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/20 : FAILURE in 3h 05m 39s","accounts_in_message":[],"_revision_number":20},{"id":"31cd3ca3af76a7f06dee044c6ae5665b63d6ac8e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 03:18:12.000000000","message":"Patch Set 20:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33160 : FAILURE in 3h 53m 47s","accounts_in_message":[],"_revision_number":20},{"id":"ab5ff4da393760249865483c2af960a8aa30d596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:23:35.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/81/639581/20/check/grenade-py3/eb74cea/ : SUCCESS in 55m 07s\n- tempest-full-py3 http://logs.openstack.org/81/639581/20/check/tempest-full-py3/9689bbd/ : SUCCESS in 1h 26m 52s\n- openstack-tox-cover http://logs.openstack.org/81/639581/20/check/openstack-tox-cover/f053105/ : FAILURE in 13m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/20/check/openstack-tox-lower-constraints/686cf9b/ : FAILURE in 18m 17s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/20/check/openstack-tox-pep8/dfa8c17/ : SUCCESS in 10m 17s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/20/check/openstack-tox-py27/facece7/ : FAILURE in 11m 24s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/20/check/openstack-tox-py36/8cd1de5/ : FAILURE in 11m 46s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/20/check/openstack-tox-py37/5c50ae0/ : FAILURE in 11m 07s\n- openstack-tox-docs http://logs.openstack.org/81/639581/20/check/openstack-tox-docs/e83386a/html/ : SUCCESS in 5m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/20/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/03f6739/ : FAILURE in 36m 46s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/20/check/devstack-plugin-ceph-tempest/bcd2e86/ : SUCCESS in 1h 31m 07s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/20/check/neutron-grenade-multinode/9d1b40e/ : SUCCESS in 1h 07m 41s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/20/check/neutron-tempest-linuxbridge/51d9c66/ : SUCCESS in 1h 32m 19s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/20/check/nova-grenade-live-migration/0435f72/ : SUCCESS in 1h 05m 28s\n- nova-live-migration http://logs.openstack.org/81/639581/20/check/nova-live-migration/4796c21/ : SUCCESS in 50m 41s\n- nova-multi-cell http://logs.openstack.org/81/639581/20/check/nova-multi-cell/2ba1bc3/ : SUCCESS in 1h 33m 08s (non-voting)\n- nova-next http://logs.openstack.org/81/639581/20/check/nova-next/73afb6b/ : SUCCESS in 1h 39m 28s\n- nova-tox-functional http://logs.openstack.org/81/639581/20/check/nova-tox-functional/4ad58d1/ : SUCCESS in 20m 36s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639581/20/check/nova-tox-functional-py36/41fda5e/ : SUCCESS in 18m 39s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/20/check/tempest-slow-py3/e9fd405/ : SUCCESS in 2h 40m 52s","accounts_in_message":[],"_revision_number":20},{"id":"485c907a564f57d080902814fafdefa9136fea64","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-26 14:32:23.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"010415dac87eb5de920b77687c9be8f80ae21f61","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-26 16:57:56.000000000","message":"Patch Set 21:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/21/check/tempest-dsvm-full-xenial/624c5e5/ : SUCCESS in 1h 31m 37s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/21/check/tempest-dsvm-full-xenial-py3/26809b7/ : SUCCESS in 1h 33m 44s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/81/639581/21/check/grenade-dsvm-xenial/6a3f9c4/ : FAILURE in 1h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ded3f9300b71d0c9283804e1ef10f70a63aad89e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-26 17:59:37.000000000","message":"Patch Set 21:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33201 : SUCCESS in 2h 03m 55s","accounts_in_message":[],"_revision_number":21},{"id":"e8570b45af557d795e2fbc1b2088374808f00813","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-26 18:25:01.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/21/check-vote/ext-nova-zuul/580d4b3 : FAILURE in 51m 45s","accounts_in_message":[],"_revision_number":21},{"id":"cc92ec7fb5b4089278cdd04a876cda80eda0aec9","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 19:37:40.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/21/check/check-dsvm-tempest-vz7-exe-minimal/a48625e : FAILURE in 10m 00s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"6375e86f6b590bdd8655836f330358e582d84e1f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-26 19:41:15.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/639581/21 : FAILURE in 2h 36m 20s","accounts_in_message":[],"_revision_number":21},{"id":"203f4dda72f20f62f3475b402c4858f9b78e6623","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-26 20:31:25.000000000","message":"Patch Set 21:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/81/639581/21/check/nova-out-of-tree-pvm/7ec5fcc : FAILURE in 5h 10m 39s\n- nova-in-tree-pvm http://184.172.12.213/81/639581/21/check/nova-in-tree-pvm/e858a8c : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":21},{"id":"f42e25f733ec48b968d24767bb2ac8f0038505ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:35:35.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/21/check/grenade-py3/36f4a59/ : SUCCESS in 1h 01m 04s\n- tempest-full-py3 http://logs.openstack.org/81/639581/21/check/tempest-full-py3/ce0c52a/ : SUCCESS in 1h 41m 56s\n- openstack-tox-cover http://logs.openstack.org/81/639581/21/check/openstack-tox-cover/9b64d33/cover/ : SUCCESS in 14m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/21/check/openstack-tox-lower-constraints/19fe303/ : SUCCESS in 16m 19s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/21/check/openstack-tox-pep8/19319ff/ : SUCCESS in 9m 19s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/21/check/openstack-tox-py27/8ed9a98/ : SUCCESS in 13m 12s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/21/check/openstack-tox-py36/358ed08/ : SUCCESS in 11m 19s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/21/check/openstack-tox-py37/3832e8c/ : SUCCESS in 9m 33s\n- openstack-tox-docs http://logs.openstack.org/81/639581/21/check/openstack-tox-docs/83467b3/html/ : SUCCESS in 7m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/21/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/18abf1c/ : SUCCESS in 53m 03s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/21/check/devstack-plugin-ceph-tempest/230cf4d/ : SUCCESS in 1h 22m 00s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/21/check/neutron-grenade-multinode/4e79866/ : SUCCESS in 1h 35m 21s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/21/check/neutron-tempest-linuxbridge/8922a4b/ : SUCCESS in 1h 50m 30s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/21/check/nova-grenade-live-migration/6c29cad/ : SUCCESS in 1h 12m 26s\n- nova-live-migration http://logs.openstack.org/81/639581/21/check/nova-live-migration/2c03fd2/ : SUCCESS in 49m 04s\n- nova-multi-cell http://logs.openstack.org/81/639581/21/check/nova-multi-cell/41a96d7/ : SUCCESS in 1h 22m 01s (non-voting)\n- nova-next http://logs.openstack.org/81/639581/21/check/nova-next/12fea1b/ : SUCCESS in 1h 21m 38s\n- nova-tox-functional http://logs.openstack.org/81/639581/21/check/nova-tox-functional/66b0c43/ : SUCCESS in 19m 06s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639581/21/check/nova-tox-functional-py36/47e9f62/ : SUCCESS in 19m 13s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/21/check/tempest-slow-py3/f2cd1c3/ : SUCCESS in 2h 53m 07s","accounts_in_message":[],"_revision_number":21},{"id":"452eb1e52561e699537afa79be8262306e845010","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 03:38:06.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/21/check/EMC_VxFlexOS_NOVA/0996355/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":21},{"id":"3dff412946dd20d2c8a1ea519f6ec811cbeb4a1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-27 07:33:52.000000000","message":"Patch Set 21:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-21 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":21},{"id":"8508012f3cbb17b6bc17e4bada7dcc1a1317a28b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 17:31:35.000000000","message":"Patch Set 21:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/21/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/b032a1c : SUCCESS in 2h 19m 20s","accounts_in_message":[],"_revision_number":21},{"id":"41293bdb1c2cd595530a033f355125b924122a8d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-31 22:13:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"0bb2fc96044f98218234368890506370cbd91c8d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-07-31 22:24:07.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"c82a9b78507a9f1f7c5149f727504f4f3c9fc14e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-01 00:03:15.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/22/check/tempest-dsvm-full-xenial/4dba00f/ : SUCCESS in 1h 32m 42s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/22/check/tempest-dsvm-full-xenial-py3/49d7f23/ : SUCCESS in 1h 28m 27s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/22/check/grenade-dsvm-xenial/549cee0/ : SUCCESS in 1h 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"60064613bb94584f5b609bef13324dffc5c24079","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-01 00:03:44.000000000","message":"Patch Set 22:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-34552 : SUCCESS in 1h 41m 48s","accounts_in_message":[],"_revision_number":22},{"id":"3b4432f8dd0729030073f49b9b0e975d392f16e3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-01 01:21:38.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/22/check/check-dsvm-tempest-vz7-exe-minimal/e8c733d : FAILURE in 11m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"f3822320edfee0e4cb1180ed44e6ae7d8a53d6cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:31:25.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/81/639581/22/check/grenade-py3/23d86cb/ : SUCCESS in 58m 20s\n- tempest-integrated-compute http://logs.openstack.org/81/639581/22/check/tempest-integrated-compute/8fa72df/ : SUCCESS in 1h 26m 11s\n- openstack-tox-cover http://logs.openstack.org/81/639581/22/check/openstack-tox-cover/dcb551b/cover/ : SUCCESS in 20m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/639581/22/check/openstack-tox-lower-constraints/e02157b/ : SUCCESS in 25m 20s\n- openstack-tox-pep8 http://logs.openstack.org/81/639581/22/check/openstack-tox-pep8/8349fc6/ : SUCCESS in 11m 01s\n- openstack-tox-py27 http://logs.openstack.org/81/639581/22/check/openstack-tox-py27/48393c1/ : SUCCESS in 17m 51s\n- openstack-tox-py36 http://logs.openstack.org/81/639581/22/check/openstack-tox-py36/4a29851/ : SUCCESS in 17m 39s\n- openstack-tox-py37 http://logs.openstack.org/81/639581/22/check/openstack-tox-py37/33ab93d/ : SUCCESS in 18m 47s\n- openstack-tox-docs http://logs.openstack.org/81/639581/22/check/openstack-tox-docs/048e218/html/ : SUCCESS in 8m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/81/639581/22/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/dee3ce6/ : SUCCESS in 58m 31s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/81/639581/22/check/devstack-plugin-ceph-tempest/a0bc795/ : SUCCESS in 1h 14m 50s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/81/639581/22/check/neutron-grenade-multinode/b8550fe/ : SUCCESS in 1h 16m 12s\n- neutron-tempest-linuxbridge http://logs.openstack.org/81/639581/22/check/neutron-tempest-linuxbridge/593acb8/ : SUCCESS in 1h 30m 36s\n- nova-grenade-live-migration http://logs.openstack.org/81/639581/22/check/nova-grenade-live-migration/b7fd437/ : SUCCESS in 1h 19m 13s\n- nova-live-migration http://logs.openstack.org/81/639581/22/check/nova-live-migration/33379c4/ : SUCCESS in 45m 13s\n- nova-multi-cell https://logs.opendev.org/81/639581/22/check/nova-multi-cell/3ec5059/ : SUCCESS in 1h 34m 01s (non-voting)\n- nova-next http://logs.openstack.org/81/639581/22/check/nova-next/a7e9a7d/ : SUCCESS in 1h 21m 05s\n- nova-tox-functional http://logs.openstack.org/81/639581/22/check/nova-tox-functional/cdc3057/ : SUCCESS in 24m 36s\n- nova-tox-functional-py36 http://logs.openstack.org/81/639581/22/check/nova-tox-functional-py36/3b9bbca/ : SUCCESS in 25m 05s\n- tempest-slow-py3 http://logs.openstack.org/81/639581/22/check/tempest-slow-py3/c5b5bcd/ : SUCCESS in 2h 13m 39s\n- tempest-ipv6-only http://logs.openstack.org/81/639581/22/check/tempest-ipv6-only/98ec8ac/ : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":22},{"id":"db127ceeae667b4a9dc6312b2d9b49df8d4bdbf4","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-01 01:45:55.000000000","message":"Patch Set 22:\n\n* pci-test http://52.27.155.124/pci/639581/22 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"ec4761e32fb64a0a0b9f2c2c05cf65cfd3e5615e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-01 04:24:54.000000000","message":"Patch Set 22:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-22 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":22},{"id":"6f17ba6f72034814be6e1fa6a71433328348465d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-08-01 05:03:16.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/81/639581/22/check-vote/ext-nova-zuul/a6ab3b0 : FAILURE in 1h 00m 36s","accounts_in_message":[],"_revision_number":22},{"id":"2aa6ad78698e702dee255f656f3a4b14118d592a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-08-01 09:07:21.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/81/639581/22/check-nova/Nova-ML2-Sriov/e38ebc9 : SUCCESS in 1h 25m 33s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"2393cd04c34920cee475d841596e4357052509ec","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:27:37.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"316d16f384b1c0b36118f2042f2c271fe1c7ceee","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:36:05.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"06989d33a7626d699ec9721adbfad49f49c8e056","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 10:37:51.000000000","message":"Patch Set 23:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/23/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1dab91f : NOT_REGISTERED","accounts_in_message":[],"_revision_number":23},{"id":"49c3276a942a3ff796d15d8dac0d540127356b42","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-16 10:42:07.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"5c36221a17f12013fcf63246d8a50c8165611968","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-08-16 12:21:52.000000000","message":"Patch Set 23:\n\n* pci-test http://52.27.155.124/pci/639581/23 : SUCCESS","accounts_in_message":[],"_revision_number":23},{"id":"a95c71f1ef2e7845b5304f06ba49687906e8c84b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-16 12:24:59.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35280 : FAILURE in 1h 47m 47s","accounts_in_message":[],"_revision_number":23},{"id":"ba9d27d1c9aa242484404e45c71e6f0104b916ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:02:11.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/92755d84f8cf41a4afb3d35b493a7cdc : SUCCESS in 1h 06m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f6ff1caf47bf4f50bbd0e9d98b57fcc0 : SUCCESS in 1h 22m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/611227ad56004781aa788422dbe2f6e4 : SUCCESS in 19m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cae2be3e3744302967e02ecdaa17af1 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086d6b2aa42146b288574fdb95a56e50 : SUCCESS in 11m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5fad4f498c04a5e9ac4cdfecbd21655 : SUCCESS in 19m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab711ab9a85b4fac84152b2add68c166 : SUCCESS in 16m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81096ca4489b4a94ad5465f2ee29c99b : SUCCESS in 21m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f60b0663c143cd862e80923b0f7d57 : SUCCESS in 8m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bb5cea3fc47b4ef68d5ea2482415fbb0 : SUCCESS in 56m 22s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fe12e0708ab4bb185934a906c61359b : SUCCESS in 1h 14m 34s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b95af9c9a3d24dfeb4408d9f65c448db : SUCCESS in 1h 02m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5697bf7bb1864262a8287e2173e6693b : SUCCESS in 1h 19m 13s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/d0e951d0a90644d5aa33b60bf07b3788 : SUCCESS in 1h 09m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a607fe673434df9a331d172eef2d7ec : SUCCESS in 48m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b350132cbcad4157885e2086248ead1b : SUCCESS in 1h 29m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/81c3772b0253468ba56a9dfa9d6c4608 : SUCCESS in 1h 21m 43s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/65ddf7523d52480facd0bd9b276f86f2 : SUCCESS in 20m 14s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f84d4c6a85264737a2e752a290da19ce : SUCCESS in 19m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a46b822f9e744380a26e5a1abbe9a600 : SUCCESS in 2h 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7d72c79f9284dabbe4111b1520424cb : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":23},{"id":"29b9a09b4bf63e3d2f2d18aeb4c6f282e77eb51a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-16 13:03:25.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/23/check/EMC_VxFlexOS_NOVA/24303bd/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"93f76d48e51c087e45e9d941e5d310ca857c01aa","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-16 19:07:05.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/23 : SUCCESS in 2h 37m 54s","accounts_in_message":[],"_revision_number":23},{"id":"fd6d2548cfd587251026edebdfd683160922e7ef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-16 23:39:30.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"ba3c7f4579a75dd21d55f6ee63a2969fa62f5de5","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-17 03:54:06.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/23/check/check-dsvm-tempest-vz7-exe-minimal/c66b3d6 : FAILURE in 11m 05s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"5f39ffc8a7898d46045542f0dbe9ec1e2e5062a1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-20 11:34:45.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"08d2315b066db3cbf9221b212b07579c65192612","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 11:37:53.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"21e7e3b200a8e8f0b27268afae1409f359b0c02d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-20 13:50:58.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35420 : FAILURE in 2h 08m 49s","accounts_in_message":[],"_revision_number":24},{"id":"3ee32e6413eae7af2aa797f65349c553d0845381","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-20 14:07:23.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/24 : SUCCESS in 2h 26m 25s","accounts_in_message":[],"_revision_number":24},{"id":"8cb540533c181392a1be53cf8c368d6d678fa4b7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-20 14:15:34.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial/30b007c/ : SUCCESS in 1h 58m 44s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial-py3/fe96912/ : FAILURE in 2h 05m 25s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/grenade-dsvm-xenial/b0ec037/ : FAILURE in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"275c2c1f6a9d1bb85cb5640ee4367bc64ec4bbdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:19:34.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/684553f42cb44d3e9a23d1c5e3935e01 : SUCCESS in 1h 02m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c2c4eba390ee46ae9dda5758cfaba5c2 : SUCCESS in 1h 23m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7b00d447d6f4bc7ba0282c5f9e0dd5e : SUCCESS in 16m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f44db0ac30d4d0eb71503b2158e76cf : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20417677848a485baeecb845c9383b6a : SUCCESS in 10m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1e53b52f77e434fa25aca4d189129a4 : SUCCESS in 13m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64cc57982e104abf9260a4fcfbd1a43a : SUCCESS in 12m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/103b31a7c34f44b4a21ca0db757968af : SUCCESS in 16m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a2945e0d414c30bfb56840bda12474 : SUCCESS in 7m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c09ea1cfe01b4c58a2b351ce9814fccf : SUCCESS in 57m 29s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63785d96d26f44b49ba0fc289a794edf : FAILURE in 1h 30m 30s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29d19487b6b04720b9955f143685d73f : SUCCESS in 1h 07m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/426f8f3635ee46469f040d822c002463 : SUCCESS in 1h 17m 40s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/38e84d319e36462aa53a2c67753907f0 : SUCCESS in 1h 08m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2161c38220aa45b081067d8819bd049c : FAILURE in 47m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b0a46d3848fe4f6aa484c67941586326 : SUCCESS in 1h 32m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/69d893bba17d4474b113d243a49ff3b6 : SUCCESS in 1h 19m 58s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/a018d5f15afd4c19934c0a8157698fb3 : SUCCESS in 21m 23s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f17c7d1cacf24f5891dc74ba7cefff5c : SUCCESS in 18m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e30f07b45bed43a287d9e8a8b99bea31 : SUCCESS in 2h 20m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a4e41c974894f97b060e7a709ccac8b : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":24},{"id":"69dcc62086ce7d8713a190dde8568cf89473e36a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 14:34:04.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/81/639581/24/check/check-dsvm-tempest-vz7-exe-minimal/c261674 : FAILURE in 10m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"95a3ad42e10bf99a83186c57e8d942074904a9e9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-20 14:48:31.000000000","message":"Patch Set 24: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"b4c307a78f84a7a827a1b1efec00a5f9ec25773c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 14:51:01.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"a630443c943607f09268795c184bc2a7bc1c3113","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 16:07:20.000000000","message":"Patch Set 24:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/639581/24/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ed2f40b : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":24},{"id":"6cbfb882cfc05dbfa4eb7e7cc68be2e7ec4c63f2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-20 16:13:05.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35439 : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":24},{"id":"34f7aa21373d2b81e837667a2d10cfba347513cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:13:41.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01a025f2a30d4b76a458a5591f2366ab : SUCCESS in 1h 00m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1892e7fc62ed4edab81210a514b3f12a : SUCCESS in 1h 33m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e8e2e33f8ee44c29557ba73d502a027 : SUCCESS in 16m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c7861d467d04f528b8278ab090e4233 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8dae60abb6d419dafcdb549076b9756 : SUCCESS in 11m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c908e838b0345949a60dce1d488537a : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6554d100b9c472c94aeb65a4b55e2b3 : SUCCESS in 12m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f756974ea2134f22a3808b86aa0ece14 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e21dfcd86c4cfa8035ef93b1fc9016 : SUCCESS in 8m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5627c87bb3ff47e6b03d26974e96e682 : SUCCESS in 55m 26s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e842fedbee024444a0266a663d811802 : SUCCESS in 1h 07m 39s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0de2499d65446ba8249ff6809d4f198 : SUCCESS in 1h 13m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0a2a2f25d1d4b6c9febffc34f7fef0e : SUCCESS in 1h 17m 43s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/ef32e026b7ce483f89188edcd19df68f : SUCCESS in 1h 12m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a1e27e5b60c9462aae159ab6ba8344ad : SUCCESS in 53m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3a517d91b1a24446a3c87cc1ede2f8af : SUCCESS in 1h 34m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/868f094f09104730bcfa9e263945cca1 : SUCCESS in 1h 27m 53s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/abd424d3b9df4fb68a5c5516901ec373 : SUCCESS in 20m 10s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cf595a8b226471582c479c2ccd6a82a : SUCCESS in 19m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43dc5b9fd3d34240b4c6c65f9beaf08c : SUCCESS in 2h 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac57dc61854496f98ee47f1cf4fa2b7 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":24},{"id":"cd7699d1eb9f6a606d274ca8ee673379b7520bca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-20 17:14:33.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/24/check/EMC_VxFlexOS_NOVA/e26df35/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":24},{"id":"6b15a40003de17d1c724680704f8c62060523133","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-20 17:28:06.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial/81edffa/ : SUCCESS in 1h 47m 46s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial-py3/cdd35d6/ : SUCCESS in 2h 00m 43s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/grenade-dsvm-xenial/89f3ca4/ : SUCCESS in 2h 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2b5217535b276033a34b1b0d37feafc7c8287501","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 19:15:36.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"5ef17098fdec28ec73e913cf9e94dc9f0bdb729e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-20 19:30:52.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/24 : SUCCESS in 2h 19m 48s","accounts_in_message":[],"_revision_number":24},{"id":"1ebbb27d1044c4d356c4ea563f5bd09abb96d25b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-20 23:15:51.000000000","message":"Patch Set 24:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-24 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":24},{"id":"d780ff7bb04b0f98375d5ebbb44ec661211d05e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 01:02:52.000000000","message":"Patch Set 24: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":24},{"id":"f8c70753dd3b90a8c0a9471f620153149c78c7a7","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-08-21 07:01:29.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"47f7d6cfeed6721a24898c81874f8c90c289be32","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 07:03:10.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1a602644015b07b44ec48aa62c25398909dc5a4b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-21 08:16:51.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35507 : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":24},{"id":"aff313ef49dbc3b20f2a1d0532e19c9874a9d2dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 10:20:15.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/937bb3039360417c88d57b428dc5d9fc : SUCCESS in 1h 14m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b2f302c519fe4357a2fb940655c9fc1f : SUCCESS in 1h 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce82ef134cc487a9a8e9a5ad03d0b2c : SUCCESS in 19m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6f94a44ec044a188ae33edc310c3bb8 : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e839d3ac95c471b8792d61542c66540 : SUCCESS in 11m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad20d7990ad141fcb522271ac9b8d26f : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1484c764f60348d29392f1fa75384e4e : SUCCESS in 14m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a4a30c7a4b54f14b49a98b69d00d096 : SUCCESS in 14m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f814e61025044d2683e44a7c694b2d09 : SUCCESS in 9m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1fcc432d1b97459a9f43b951e6e3ee8c : SUCCESS in 52m 27s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86fc6292508f45a784518bde52237085 : FAILURE in 1h 21m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c1d91256b59453d91d04f667ed2b7c8 : SUCCESS in 1h 13m 29s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f8a74465082840108a6b6d7f9ba9a8fb : SUCCESS in 1h 20m 22s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/847ca9428ce1436895de442ded3c123e : SUCCESS in 1h 10m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/139ee5f916d442a9be96e46d5f40e750 : SUCCESS in 1h 12m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2f4caf00800b48a1b54a49d23c0ffaf6 : SUCCESS in 1h 26m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/e7fc81af57fa401698ce554eb65b8f43 : SUCCESS in 1h 33m 56s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/2b9ed1aded054a52862be0e39949ac00 : SUCCESS in 19m 26s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/372b35bdad0b42fd89d58619d77bf40a : SUCCESS in 17m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e1f815147f34a7780bd3dd3963345ab : SUCCESS in 2h 30m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd85af97ae9a46eab8a299590a02d907 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":24},{"id":"4a7c2acc5df5f7b46464747cc883fb2fdfcfd121","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-21 10:21:02.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/81/639581/24/check/EMC_VxFlexOS_NOVA/b6455d5/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":24},{"id":"bbd282cc1e26691c0b7a193c1ad6d59d0e19cbfb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-21 11:04:45.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial/e70f20e/ : SUCCESS in 1h 59m 25s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/tempest-dsvm-full-xenial-py3/dc47a34/ : SUCCESS in 2h 05m 42s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/81/639581/24/check/grenade-dsvm-xenial/0970f6c/ : SUCCESS in 1h 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"e3d02ba68b4d74017ea7828969bad726ea3d37bb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-21 12:38:50.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/639581/24 : SUCCESS in 2h 21m 05s","accounts_in_message":[],"_revision_number":24},{"id":"8d8b6635b6f6a4490ed6f2a08b20c48cc74a321f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-21 18:32:43.000000000","message":"Patch Set 24:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-81-639581-24 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":24},{"id":"4dfff7b88acaf3fc24d0207d6e243292a0536ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 00:52:08.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"57d1c57154a6aaa0715340111a6e4027d6ea79a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:12:59.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2141e48f30d438689ef09c86fbd6210 : SUCCESS in 58m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/81fc8d7585134191afaf3dd4c39ebc2d : SUCCESS in 1h 26m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3acbcf5f2c3242b58b92844bbbfd9ded : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedeffe0cedb4dffbb6b76a1e11bc1c7 : SUCCESS in 11m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/224655d5af7f42d3b98f67697ad967b0 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63e308606394a9a8d56346830cf9dc2 : SUCCESS in 11m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/510d00059a5e4d2b8852ea077f00ff7e : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5336747c22430a94fa240d67e8f053 : SUCCESS in 8m 58s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/ef09022e35ac447094c6e45a0e2224b6 : SUCCESS in 1h 07m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b7f0ca5629d949fcb23878822ff97d1f : SUCCESS in 55m 44s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/08ac3c6cc2234ac1badc7a873309536b : SUCCESS in 20m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d409bb6f057b40bda55cd20621656ca3 : SUCCESS in 17m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/248dde15363c41af86a682d2c69aa1ae : SUCCESS in 1h 32m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/cbb6c2ffb3504ef7bbc0376aec7cb809 : SUCCESS in 1h 19m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a158d5e122e4591a1a3fd68cdab474c : SUCCESS in 2h 20m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/400465ecb1114b638466ec9fc0af6376 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":24},{"id":"cde96d628e0ce2b9a807eff4279d935641094925","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:13:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"fb84430fd57a32e8184b350502d163a32f1c2eb1","revisions":{"e873d65a3880ecc90e98c496d0b8bdf34f35c89d":{"kind":"REWORK","_number":1,"created":"2019-02-27 06:59:29.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/1"}}},"commit":{"parents":[{"commit":"c768c0bd84dbcbd5a9991c9bef9e93a2528e13ab","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c768c0bd84dbcbd5a9991c9bef9e93a2528e13ab"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 04:40:45.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e873d65a3880ecc90e98c496d0b8bdf34f35c89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e873d65a3880ecc90e98c496d0b8bdf34f35c89d"}]},"branch":"refs/heads/master"},"9ccdac4e86278e8ee8b0ad72d558c808fa383f78":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-03-05 22:35:50.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/2"}}},"commit":{"parents":[{"commit":"d9e10a86f364add70ae3894013226d162a57e8a3","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9e10a86f364add70ae3894013226d162a57e8a3"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-05 22:18:11.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ccdac4e86278e8ee8b0ad72d558c808fa383f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ccdac4e86278e8ee8b0ad72d558c808fa383f78"}]},"branch":"refs/heads/master"},"90bd3b30a4e9c46da414f947fbe25c92c99fb06e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-06 19:09:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/639581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/3"}}},"commit":{"parents":[{"commit":"5e5177c74f294c3bbcdd087c9a57454460931259","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e5177c74f294c3bbcdd087c9a57454460931259"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2019-03-06 19:51:05.000000000","tz":-480},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90bd3b30a4e9c46da414f947fbe25c92c99fb06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90bd3b30a4e9c46da414f947fbe25c92c99fb06e"}]},"branch":"refs/heads/master"},"f4c8ef2cc8b98f9652c4ac3a66aba7225d79a4ce":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-03-07 12:06:34.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/4"}}},"commit":{"parents":[{"commit":"d82cf9d40722c690f30e4220d2c90958c42792d7","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d82cf9d40722c690f30e4220d2c90958c42792d7"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-07 12:05:59.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4c8ef2cc8b98f9652c4ac3a66aba7225d79a4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f4c8ef2cc8b98f9652c4ac3a66aba7225d79a4ce"}]},"branch":"refs/heads/master"},"9417a8ca5dc402708bcf8aa7df71f0de082a9c36":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-08 05:17:40.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/5"}}},"commit":{"parents":[{"commit":"90989cc95b854caf6fc6a2b98f7beac99b18c5a2","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90989cc95b854caf6fc6a2b98f7beac99b18c5a2"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-08 00:12:30.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9417a8ca5dc402708bcf8aa7df71f0de082a9c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9417a8ca5dc402708bcf8aa7df71f0de082a9c36"}]},"branch":"refs/heads/master"},"4a74d10a0f469eb5e33fdbfab9c4ead212dae89f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-18 05:44:49.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/6"}}},"commit":{"parents":[{"commit":"ceca75f699088faf8cade42724e26225b8220e20","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ceca75f699088faf8cade42724e26225b8220e20"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-17 23:19:10.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a74d10a0f469eb5e33fdbfab9c4ead212dae89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a74d10a0f469eb5e33fdbfab9c4ead212dae89f"}]},"branch":"refs/heads/master"},"66ddc2b6246f36852b2ad365f028a4f2685a1b79":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-21 09:37:18.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/7"}}},"commit":{"parents":[{"commit":"0b72090fe15791920eed2684e76056a74c52d6dd","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b72090fe15791920eed2684e76056a74c52d6dd"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-21 09:37:02.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66ddc2b6246f36852b2ad365f028a4f2685a1b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/66ddc2b6246f36852b2ad365f028a4f2685a1b79"}]},"branch":"refs/heads/master"},"75b61eebc370149a0fc355d5d5e6a61fc62494b1":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-03-27 04:38:51.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/8"}}},"commit":{"parents":[{"commit":"d3e37d64628dee864bfc93091a08421f035dca74","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d3e37d64628dee864bfc93091a08421f035dca74"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-26 23:16:53.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75b61eebc370149a0fc355d5d5e6a61fc62494b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75b61eebc370149a0fc355d5d5e6a61fc62494b1"}]},"branch":"refs/heads/master"},"8c1c1e9913b4c2542e17178f3a96936532f3773f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-03-29 04:13:05.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/9"}}},"commit":{"parents":[{"commit":"f0956134acbf2269be91df02273b6e6b9d75c9f4","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f0956134acbf2269be91df02273b6e6b9d75c9f4"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 04:10:28.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c1c1e9913b4c2542e17178f3a96936532f3773f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c1c1e9913b4c2542e17178f3a96936532f3773f"}]},"branch":"refs/heads/master"},"dd1af7ee1ca0e9ad7209c594ec7e890afef92fd2":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-29 06:09:26.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/10"}}},"commit":{"parents":[{"commit":"34cf589b3f6e5278e68b8b1ce1b07e3313b99059","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34cf589b3f6e5278e68b8b1ce1b07e3313b99059"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-29 06:04:21.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd1af7ee1ca0e9ad7209c594ec7e890afef92fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd1af7ee1ca0e9ad7209c594ec7e890afef92fd2"}]},"branch":"refs/heads/master"},"87e7629568d6ac2cf39dbdb5d5695427f52e4d95":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-03-29 20:45:30.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/639581/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/11"}}},"commit":{"parents":[{"commit":"3d7cf307adbe6baa9fa5e1529b831022d5c68340","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d7cf307adbe6baa9fa5e1529b831022d5c68340"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-03-29 20:41:16.000000000","tz":-300},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87e7629568d6ac2cf39dbdb5d5695427f52e4d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87e7629568d6ac2cf39dbdb5d5695427f52e4d95"}]},"branch":"refs/heads/master"},"025371802503fe566ac6cf63734289a90f18438f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-04-01 21:18:06.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/12"}}},"commit":{"parents":[{"commit":"d33376660c212becb32282355005dfdb63d77043","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d33376660c212becb32282355005dfdb63d77043"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-01 21:15:55.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/025371802503fe566ac6cf63734289a90f18438f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/025371802503fe566ac6cf63734289a90f18438f"}]},"branch":"refs/heads/master"},"ba8dd87e04ba315f50188f9d7601a72b7669f73a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-04-02 10:13:10.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/13"}}},"commit":{"parents":[{"commit":"512966c58abd122ab0998d18bc9314938c2890d3","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/512966c58abd122ab0998d18bc9314938c2890d3"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-02 10:11:06.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba8dd87e04ba315f50188f9d7601a72b7669f73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba8dd87e04ba315f50188f9d7601a72b7669f73a"}]},"branch":"refs/heads/master"},"8eb91ae026e976ce95fbff6cf2056791fcde8dd4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-04 08:44:44.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/14"}}},"commit":{"parents":[{"commit":"b2923cba522c4453919e09738202308208b85ae2","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2923cba522c4453919e09738202308208b85ae2"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 08:43:47.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8eb91ae026e976ce95fbff6cf2056791fcde8dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8eb91ae026e976ce95fbff6cf2056791fcde8dd4"}]},"branch":"refs/heads/master"},"867435bcb08d21bde9f02c90011de5e7b1df4295":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-04-04 20:44:48.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/15"}}},"commit":{"parents":[{"commit":"8156c444084828070dd9dca7b98631e6cb128bae","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8156c444084828070dd9dca7b98631e6cb128bae"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-04 20:44:03.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/867435bcb08d21bde9f02c90011de5e7b1df4295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/867435bcb08d21bde9f02c90011de5e7b1df4295"}]},"branch":"refs/heads/master"},"8488eeb76e11c0c77c74b1d7674dd81b82e7d94b":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-12 06:57:52.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/16"}}},"commit":{"parents":[{"commit":"5d556206b371600c00eb666003e59c3769647849","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d556206b371600c00eb666003e59c3769647849"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-11 22:22:09.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8488eeb76e11c0c77c74b1d7674dd81b82e7d94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8488eeb76e11c0c77c74b1d7674dd81b82e7d94b"}]},"branch":"refs/heads/master"},"479cc18ba937a17df3bd2adb9d0930bf72a9a02b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-04-24 01:44:56.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/81/639581/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/17"}}},"commit":{"parents":[{"commit":"d7b333ede4949b1476ddcf71152616c11cf7a722","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d7b333ede4949b1476ddcf71152616c11cf7a722"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-23 23:16:31.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/479cc18ba937a17df3bd2adb9d0930bf72a9a02b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/479cc18ba937a17df3bd2adb9d0930bf72a9a02b"}]},"branch":"refs/heads/master"},"430c61d3bb7e8206ed9fbc1cce6400995dfabdcd":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-06-25 16:30:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639581/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/18"}}},"commit":{"parents":[{"commit":"fb6799f459a5b7affbbfcad6020ef0d42408071c","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb6799f459a5b7affbbfcad6020ef0d42408071c"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:24:50.000000000","tz":60},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/430c61d3bb7e8206ed9fbc1cce6400995dfabdcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/430c61d3bb7e8206ed9fbc1cce6400995dfabdcd"}]},"branch":"refs/heads/master"},"5393669e561fd4287fedff8173ed54c21b1f9064":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-06-25 16:51:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639581/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/20"}}},"commit":{"parents":[{"commit":"a7a8528e9bc664e37ef7a81adc5d6f04f9261008","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7a8528e9bc664e37ef7a81adc5d6f04f9261008"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-25 16:51:20.000000000","tz":60},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5393669e561fd4287fedff8173ed54c21b1f9064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5393669e561fd4287fedff8173ed54c21b1f9064"}]},"branch":"refs/heads/master"},"2463ee5f6d5e641d46bbcee0c9446fe20c4b3eb9":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-06-26 14:32:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639581/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/21"}}},"commit":{"parents":[{"commit":"fa45236c087b1cc0fe96cbb7e41cccce84c824aa","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa45236c087b1cc0fe96cbb7e41cccce84c824aa"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-26 14:30:20.000000000","tz":60},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2463ee5f6d5e641d46bbcee0c9446fe20c4b3eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2463ee5f6d5e641d46bbcee0c9446fe20c4b3eb9"}]},"branch":"refs/heads/master"},"9d15b6962cf2c9cfc0131aeb792fbf740c248712":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-07-31 22:13:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/81/639581/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/22"}}},"commit":{"parents":[{"commit":"69ad839b0005a1adb3da7ef9e24e6fd7be279701","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69ad839b0005a1adb3da7ef9e24e6fd7be279701"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-31 22:13:27.000000000","tz":-300},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d15b6962cf2c9cfc0131aeb792fbf740c248712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d15b6962cf2c9cfc0131aeb792fbf740c248712"}]},"branch":"refs/heads/master"},"20c81a7de152fca8996bc859df6c4f4ed9db8d02":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-08-16 10:27:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639581/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/23"}}},"commit":{"parents":[{"commit":"186fb3cf52bcc2e55f5a3e95fc99a37edd91cdcf","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/186fb3cf52bcc2e55f5a3e95fc99a37edd91cdcf"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-16 10:27:18.000000000","tz":60},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20c81a7de152fca8996bc859df6c4f4ed9db8d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/20c81a7de152fca8996bc859df6c4f4ed9db8d02"}]},"branch":"refs/heads/master"},"fb84430fd57a32e8184b350502d163a32f1c2eb1":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-08-20 11:34:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/639581/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/639581/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/639581/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/639581/24"}}},"commit":{"parents":[{"commit":"7c82342d574fbb6ff3925480fbdd5eceaf79c9f3","subject":"Move final bridge commands to privsep.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c82342d574fbb6ff3925480fbdd5eceaf79c9f3"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-02-27 03:36:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-02-27 03:36:07.000000000","tz":0},"subject":"Cleanup the _execute shim in nova/network.","message":"Cleanup the _execute shim in nova/network.\n\nOriginally for testing, it has been progressively replaced by more\nspecific mocking and can now be removed.\n\nChange-Id: I11bf62d5ef3041541459d138bedb50aaeb702ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb84430fd57a32e8184b350502d163a32f1c2eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb84430fd57a32e8184b350502d163a32f1c2eb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
