)]}'
{"id":"openstack%2Fnova~641138","triplet_id":"openstack%2Fnova~master~Id6c0a0ee41520dd974052d7cdd17ca35d688f6b0","project":"openstack/nova","branch":"master","topic":"neutronapi-migrate-instance-finish","hashtags":[],"change_id":"Id6c0a0ee41520dd974052d7cdd17ca35d688f6b0","subject":"Use errors_out_migration decorator on finish_resize","status":"MERGED","created":"2019-03-05 21:59:09.000000000","updated":"2019-07-01 18:34:24.000000000","submitted":"2019-03-29 00:14:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"641138-1553818496443-7c43b880","meta_rev_id":"89b0a7f13ac9a1a1dad5d8de38edee650738b7ce","_number":641138,"virtual_id_number":641138,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-03-28 16:30:20.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-06 14:50:00.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-28 16:35:13.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-28 15:26:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-28 15:08:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-03-29 00:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"date":"2019-03-06 06:36:49.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-28 17:52:30.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"date":"2019-03-28 16:11:40.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-03-06 17:42:28.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"date":"2019-03-28 13:55:23.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-28 14:59:53.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-28 09:49:03.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-03-05 22:52:17.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-03-28 09:49:03.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-05 22:01:25.000000000","_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-05 22:01:25.000000000","updated_by":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"reviewer":{"_account_id":9555,"name":"Matthew Booth","email":"mbooth@redhat.com","username":"MatthewBooth"},"state":"REVIEWER"},{"updated":"2019-03-05 22:52:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-06 06:36:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-06 14:50:00.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-06 17:42:28.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-03-28 09:49:03.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-03-28 13:55:23.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-28 15:08:26.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-03-28 15:26:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-28 16:11:40.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-03-28 16:30:20.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-28 16:35:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-28 17:52:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-29 00:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f24e935e7a287092bcb7eb25ff8f87dd2fbf4207","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-05 21:59:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b6c14d16e14a9408ac276abc13aa643252d6618","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-05 22:02:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96d556c912d990978919a532bfba97a09546daf6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-05 22:24:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nChange good. Please add bug (or convince me we don\u0027t need one).","accounts_in_message":[],"_revision_number":1},{"id":"fbdd7a9b83b803c490345db94c4ea612a89951fe","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-05 22:45:25.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/38/641138/1/check/nova-out-of-tree-pvm/222720f : FAILURE in 44m 01s\n- nova-in-tree-pvm http://184.172.12.213/38/641138/1/check/nova-in-tree-pvm/f83fece : FAILURE in 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a573d6759e8bbb92425986f226262960b631ee26","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-05 22:45:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90695e312995feb34a6d658623fc8e6d63ccb231","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-05 22:52:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fd6f2168e7356f5c48931139aa19756e8b3ac68","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-06 00:17:37.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-28719 : SUCCESS in 1h 56m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2ed0ab4171d6811b4626b10260e27dcb31bedacc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 01:07:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/38/641138/1/check/grenade-py3/73d77a6/ : SUCCESS in 1h 00m 32s\n- tempest-full-py3 http://logs.openstack.org/38/641138/1/check/tempest-full-py3/ea65de5/ : SUCCESS in 1h 28m 18s\n- openstack-tox-cover http://logs.openstack.org/38/641138/1/check/openstack-tox-cover/21d0df5/cover/ : SUCCESS in 18m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/641138/1/check/openstack-tox-lower-constraints/c265d66/ : SUCCESS in 11m 01s\n- openstack-tox-pep8 http://logs.openstack.org/38/641138/1/check/openstack-tox-pep8/90fc149/ : SUCCESS in 11m 37s\n- openstack-tox-py27 http://logs.openstack.org/38/641138/1/check/openstack-tox-py27/83346e6/ : SUCCESS in 13m 48s\n- openstack-tox-py35 http://logs.openstack.org/38/641138/1/check/openstack-tox-py35/2a361c4/ : SUCCESS in 12m 10s\n- openstack-tox-py36 http://logs.openstack.org/38/641138/1/check/openstack-tox-py36/49240ba/ : SUCCESS in 12m 08s\n- openstack-tox-docs http://logs.openstack.org/38/641138/1/check/openstack-tox-docs/dcb1ecf/html/ : SUCCESS in 6m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/38/641138/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6ce7739/ : SUCCESS in 1h 00m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/38/641138/1/check/devstack-plugin-ceph-tempest/9a048d6/ : SUCCESS in 2h 00m 43s (non-voting)\n- legacy-grenade-dsvm-neutron-multinode-live-migration http://logs.openstack.org/38/641138/1/check/legacy-grenade-dsvm-neutron-multinode-live-migration/0acc742/ : SUCCESS in 1h 06m 30s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/38/641138/1/check/neutron-grenade-multinode/403e7aa/ : SUCCESS in 1h 10m 41s\n- nova-live-migration http://logs.openstack.org/38/641138/1/check/nova-live-migration/4112326/ : SUCCESS in 48m 01s\n- nova-next http://logs.openstack.org/38/641138/1/check/nova-next/49e3863/ : SUCCESS in 1h 59m 15s\n- nova-tox-functional http://logs.openstack.org/38/641138/1/check/nova-tox-functional/8aaa40b/ : SUCCESS in 23m 28s\n- nova-tox-functional-py35 http://logs.openstack.org/38/641138/1/check/nova-tox-functional-py35/3be5f5e/ : SUCCESS in 16m 48s\n- tempest-multinode-full http://logs.openstack.org/38/641138/1/check/tempest-multinode-full/ba55fa8/ : SUCCESS in 1h 32m 04s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/38/641138/1/check/tempest-slow-py3/a48cc0e/ : SUCCESS in 2h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3b157900b579d6dece0d4f289ca88a655d290e1b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-06 01:08:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/38/641138/1/check/EMC_VxFlexOS_NOVA/b23575b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"80f6792bbd944871bd7da686e342d3402828e3c5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-06 01:12:05.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/641138/1 : FAILURE in 2h 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"401cddee16f76de48a98ebaad9f17fd87b33f703","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-06 02:58:43.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-38-641138-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"3183af483a1ed23735e8985dac611c077a843e39","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-06 06:36:49.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/38/641138/1/check-vote/ext-nova-zuul/f9ad46b : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c3a5a84a8425b28eca918d0a4aad98f5387ba9bc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-06 14:50:00.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/641138/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/9166eb8 : SUCCESS in 1h 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4992ce56729bcc14d0490fe3bd8d415844116f9a","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-06 17:42:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebe6805ac5464db6b9ba8d00f6ce0a79bfb81863","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-03-28 09:49:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMakes good sense.","accounts_in_message":[],"_revision_number":1},{"id":"77f293403d49b3f4c98fdf635784739251ec18de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 09:49:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"145ef4f76a17821ecc5adadaf2590306608efb97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 13:48:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"8dee47014aec532bf9e6f1343e9d9d6c0eed991a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-28 13:52:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc72c510c68eeca7874ee2a8a4697196963345d3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-28 13:55:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a305eea72b00b415f7e5295add80125b3a9610fe","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-28 14:56:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3675e69f1cff876ad9200dd05e019ab94f0723b9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-03-28 14:59:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1a2760f938c03a51afffddaf3e15edec81b99eb","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-28 15:08:26.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29629 : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a04305eb5fa80fab7cb8aa036e739d3355d34de1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-28 15:26:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/38/641138/1/check/tempest-dsvm-full-xenial/518b5f2/ : SUCCESS in 1h 15m 44s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/38/641138/1/check/tempest-dsvm-full-xenial-py3/ffffa28/ : SUCCESS in 1h 29m 45s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/38/641138/1/check/grenade-dsvm-xenial/b79bf77/ : SUCCESS in 1h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c34174692c0076e52b2acc6623af22061bdf06b6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-28 16:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/641138/1 : SUCCESS in 2h 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"89b8d08945bedcd52a44211a3468a4b749773773","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-28 16:30:20.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/38/641138/1/check/nova-out-of-tree-pvm/1e0278f : FAILURE in 2h 36m 16s\n- nova-in-tree-pvm http://184.172.12.213/38/641138/1/check/nova-in-tree-pvm/a9dae0d : FAILURE in 2h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1b7ffd565a2d31b1b9dc786326a24b71f53f03d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 16:34:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/38/641138/1/check/grenade-py3/228cfd2/ : SUCCESS in 56m 38s\n- tempest-full-py3 http://logs.openstack.org/38/641138/1/check/tempest-full-py3/727c1e2/ : SUCCESS in 1h 35m 54s\n- openstack-tox-cover http://logs.openstack.org/38/641138/1/check/openstack-tox-cover/44c60bf/cover/ : SUCCESS in 16m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/641138/1/check/openstack-tox-lower-constraints/5c9b5e8/ : SUCCESS in 12m 54s\n- openstack-tox-pep8 http://logs.openstack.org/38/641138/1/check/openstack-tox-pep8/e3eda9b/ : SUCCESS in 11m 58s\n- openstack-tox-py27 http://logs.openstack.org/38/641138/1/check/openstack-tox-py27/20d559e/ : SUCCESS in 15m 14s\n- openstack-tox-py35 http://logs.openstack.org/38/641138/1/check/openstack-tox-py35/9668008/ : SUCCESS in 12m 59s\n- openstack-tox-py36 http://logs.openstack.org/38/641138/1/check/openstack-tox-py36/2e8731d/ : SUCCESS in 19m 51s\n- openstack-tox-py37 http://logs.openstack.org/38/641138/1/check/openstack-tox-py37/cb64660/ : SUCCESS in 13m 22s\n- openstack-tox-docs http://logs.openstack.org/38/641138/1/check/openstack-tox-docs/0e3f7bb/html/ : SUCCESS in 7m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/38/641138/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b56a2bc/ : SUCCESS in 49m 14s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/38/641138/1/check/devstack-plugin-ceph-tempest/6d4d9c5/ : SUCCESS in 1h 22m 52s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/38/641138/1/check/neutron-grenade-multinode/d00ec28/ : SUCCESS in 1h 04m 48s\n- nova-grenade-live-migration http://logs.openstack.org/38/641138/1/check/nova-grenade-live-migration/69dcda2/ : SUCCESS in 1h 07m 14s\n- nova-live-migration http://logs.openstack.org/38/641138/1/check/nova-live-migration/04c403b/ : SUCCESS in 52m 27s\n- nova-next http://logs.openstack.org/38/641138/1/check/nova-next/fe8b8b5/ : SUCCESS in 1h 33m 47s\n- nova-tox-functional http://logs.openstack.org/38/641138/1/check/nova-tox-functional/83336b0/ : SUCCESS in 20m 54s\n- nova-tox-functional-py35 http://logs.openstack.org/38/641138/1/check/nova-tox-functional-py35/e1e2b4d/ : SUCCESS in 18m 10s\n- tempest-slow-py3 http://logs.openstack.org/38/641138/1/check/tempest-slow-py3/b556e93/ : SUCCESS in 2h 27m 18s","accounts_in_message":[],"_revision_number":1},{"id":"50d694273c147cf8cb389e627b242e23794f17eb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-28 16:35:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/38/641138/1/check/EMC_VxFlexOS_NOVA/5d618f2/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"38e4a931f1fdbd8baaae0f4d55684a878bb06aa0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-28 17:52:30.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-38-641138-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"8358b684d5d3fdac6df2de0d6c56076a88f575bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 21:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0ae486d936a467aec266e527ec352e363187844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 00:14:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/38/641138/1/gate/grenade-py3/efc7a04/ : SUCCESS in 54m 56s\n- tempest-full-py3 http://logs.openstack.org/38/641138/1/gate/tempest-full-py3/1459428/ : SUCCESS in 1h 41m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/641138/1/gate/openstack-tox-lower-constraints/1f3905a/ : SUCCESS in 11m 39s\n- openstack-tox-pep8 http://logs.openstack.org/38/641138/1/gate/openstack-tox-pep8/4cd582f/ : SUCCESS in 13m 34s\n- openstack-tox-py27 http://logs.openstack.org/38/641138/1/gate/openstack-tox-py27/ae0d9bf/ : SUCCESS in 13m 32s\n- openstack-tox-py35 http://logs.openstack.org/38/641138/1/gate/openstack-tox-py35/f8d8486/ : SUCCESS in 21m 49s\n- openstack-tox-py36 http://logs.openstack.org/38/641138/1/gate/openstack-tox-py36/077ec49/ : SUCCESS in 21m 25s\n- openstack-tox-py37 http://logs.openstack.org/38/641138/1/gate/openstack-tox-py37/997e55e/ : SUCCESS in 12m 40s\n- openstack-tox-docs http://logs.openstack.org/38/641138/1/gate/openstack-tox-docs/d904b32/html/ : SUCCESS in 6m 16s\n- nova-grenade-live-migration http://logs.openstack.org/38/641138/1/gate/nova-grenade-live-migration/9b029f4/ : SUCCESS in 1h 08m 21s\n- nova-live-migration http://logs.openstack.org/38/641138/1/gate/nova-live-migration/de9d8b0/ : SUCCESS in 57m 37s\n- nova-tox-functional http://logs.openstack.org/38/641138/1/gate/nova-tox-functional/1895f2e/ : SUCCESS in 21m 44s\n- nova-tox-functional-py35 http://logs.openstack.org/38/641138/1/gate/nova-tox-functional-py35/9885393/ : SUCCESS in 21m 32s\n- nova-next http://logs.openstack.org/38/641138/1/gate/nova-next/3f2bb9f/ : SUCCESS in 1h 30m 15s\n- tempest-slow-py3 http://logs.openstack.org/38/641138/1/gate/tempest-slow-py3/85dddce/ : SUCCESS in 2h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b402e4eed9b16302e88d0d05b1cddf7f2be2db39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 00:14:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8ead0a9a48615eedda1fcc36b5a3142195afd403","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-01 18:34:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08eaba7db861fb427b2ce7c3a7ac3c97887eeef3","revisions":{"08eaba7db861fb427b2ce7c3a7ac3c97887eeef3":{"kind":"REWORK","_number":1,"created":"2019-03-05 21:59:09.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/38/641138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/38/641138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/38/641138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/38/641138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/38/641138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/38/641138/1"}}},"commit":{"parents":[{"commit":"1ecdccbdaa40a080f0dc8416068fd132b661ac1e","subject":"Delete the obj_as_admin context manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ecdccbdaa40a080f0dc8416068fd132b661ac1e"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-03-05 21:52:35.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-03-05 21:54:05.000000000","tz":-300},"subject":"Use errors_out_migration decorator on finish_resize","message":"Use errors_out_migration decorator on finish_resize\n\nThis replaces the errors_out_migration_ctxt usage with the\nerrors_out_migration decorator which does the same thing. Note\nthat I9bcc1605945ddc35df2288be72e194c050b8ddd9 originally changed\nhow migration status error handling worked but at this point they\nare the same and if *anything* fails in finish_resize then the\nmigration status should be \u0027error\u0027 rather than just if part of it\nfails.\n\nChange-Id: Id6c0a0ee41520dd974052d7cdd17ca35d688f6b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08eaba7db861fb427b2ce7c3a7ac3c97887eeef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08eaba7db861fb427b2ce7c3a7ac3c97887eeef3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1ecdccbdaa40a080f0dc8416068fd132b661ac1e","is_merged_in_target_branch":true,"change_id":"I49034f87e240775cca7a4d769819dd788b86832a","change_number":635351,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
