)]}'
{"id":"openstack%2Fnova~642557","triplet_id":"openstack%2Fnova~master~Ief2c959d7c339e11b19e521de6e7d1be2f8722f8","project":"openstack/nova","branch":"master","topic":"fake-cleanup","hashtags":[],"change_id":"Ief2c959d7c339e11b19e521de6e7d1be2f8722f8","subject":"Remove fake_libvirt_utils from connection tests.","status":"MERGED","created":"2019-03-11 20:23:54.000000000","updated":"2019-05-16 21:38:34.000000000","submitted":"2019-05-16 21:38:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"642557-1558042714467-172dd170","meta_rev_id":"956e1efce261af110dbdf8ade0f84ac624332e24","_number":642557,"virtual_id_number":642557,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-13 22:43:42.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 13:53:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-05-16 21:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 00:05:12.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 14:46:05.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 10:34:56.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"date":"2019-05-14 00:16:11.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-05-13 20:27:48.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-16 19:53:21.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-05-16 19:53:21.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-11 20:24:03.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-21 15:13:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-23 02:06:36.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 20:16:35.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-13 22:43:42.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 00:05:12.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 00:16:11.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 10:34:56.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-14 13:53:11.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-05-14 14:46:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-16 19:53:21.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-05-16 21:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d85062076db5a6767da6d6451e5987c0e729d9cd","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-11 20:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e97891aeff6a6547c5b53cc2d792298d354c4e48","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-03-11 20:24:03.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/642557/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6233968a2979dc45d94b33e5ed8980a2da219330","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-11 20:25:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47733fc68f61908978c3f72495cf1acee629a6ae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-11 20:41:29.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nOn board with the idea. Unless you just hate it for some reason, I would prefer exploiting mock a little more to leave the method signatures unchanged and reduce the diff.","accounts_in_message":[],"_revision_number":1},{"id":"1461ef7fae63762a9404d87f3c9b6333a67b9272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 20:50:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/1/check/openstack-tox-cover/38896a0/cover/ : SUCCESS in 20m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/1/check/openstack-tox-lower-constraints/1843c7c/ : SUCCESS in 12m 43s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/1/check/openstack-tox-pep8/e3ddeb2/ : SUCCESS in 9m 43s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/1/check/openstack-tox-py27/87845b9/ : SUCCESS in 12m 49s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/1/check/openstack-tox-py35/4db2d42/ : SUCCESS in 13m 50s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/1/check/openstack-tox-py36/451f035/ : SUCCESS in 10m 41s\n- openstack-tox-docs http://logs.openstack.org/57/642557/1/check/openstack-tox-docs/d317443/html/ : SUCCESS in 7m 51s\n- nova-tox-functional http://logs.openstack.org/57/642557/1/check/nova-tox-functional/3d77fb6/ : SUCCESS in 21m 21s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/1/check/nova-tox-functional-py35/2a36464/ : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cda718ca91f791fb23830f297c2aa41e4fa66ee3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-11 21:24:30.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/1/check-vote/ext-nova-zuul/121b6b7 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4e7aaf4533eff0a877dc173c673dbae9408684c9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-11 21:56:16.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29072 : SUCCESS in 1h 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"de106ed6936656e1d6a81e0a6e7835295cd4b681","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-11 23:02:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642557/1 : FAILURE in 2h 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ec40fc8b6dc62d8affda42bc32d7623a8482eed2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-12 01:41:43.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"5940643b412d1bb469278bc247d6891a50ebaf8d","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-18 09:22:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d5852946a13ed378546da1844bc5f3cbaec9cde3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 09:23:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76c08339c2276caf07cd3d3ee8c2570a5ad023f5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-18 09:36:10.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29297 : FAILURE in 9m 10s","accounts_in_message":[],"_revision_number":2},{"id":"53206472a4b55013f5fead7acb10cb54581b1bcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 09:50:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/2/check/openstack-tox-cover/741585c/cover/ : SUCCESS in 16m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/2/check/openstack-tox-lower-constraints/029c1f0/ : SUCCESS in 12m 33s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/2/check/openstack-tox-pep8/defdc16/ : SUCCESS in 9m 43s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/2/check/openstack-tox-py27/22680ad/ : SUCCESS in 11m 11s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/2/check/openstack-tox-py35/f21ac9c/ : SUCCESS in 19m 30s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/2/check/openstack-tox-py36/d8045c5/ : SUCCESS in 17m 48s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/2/check/openstack-tox-py37/f06bef6/ : SUCCESS in 12m 44s\n- openstack-tox-docs http://logs.openstack.org/57/642557/2/check/openstack-tox-docs/9533bd1/html/ : SUCCESS in 6m 40s\n- nova-tox-functional http://logs.openstack.org/57/642557/2/check/nova-tox-functional/71a50dc/ : SUCCESS in 17m 43s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/2/check/nova-tox-functional-py35/6ac9de7/ : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3cd1231c64467cd8f83dff2b52ba75fc8af4e528","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 11:43:57.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/2/check-vote/ext-nova-zuul/5e10ce8 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b44eef3e9e95bdd3602b9679b484f7d26205b62c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-18 13:10:16.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642557/2 : FAILURE in 2h 50m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d899188fa09b2a8fc5736ba09b41820eb39cbac4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-18 17:10:42.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"869fef88e3761002fcc7ab2e222636f51df2163b","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-21 09:37:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"121e36c6912937751c397c7c9789178aea2d33ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 10:05:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/3/check/openstack-tox-cover/32a01f4/cover/ : SUCCESS in 16m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/3/check/openstack-tox-lower-constraints/adaf385/ : SUCCESS in 13m 23s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/3/check/openstack-tox-pep8/dd4b38e/ : SUCCESS in 11m 46s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/3/check/openstack-tox-py27/781fc75/ : SUCCESS in 14m 46s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/3/check/openstack-tox-py35/9a8dfa8/ : SUCCESS in 13m 00s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/3/check/openstack-tox-py36/4f07199/ : SUCCESS in 13m 38s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/3/check/openstack-tox-py37/4931043/ : SUCCESS in 14m 48s\n- openstack-tox-docs http://logs.openstack.org/57/642557/3/check/openstack-tox-docs/44aa5ea/html/ : SUCCESS in 7m 48s\n- nova-tox-functional http://logs.openstack.org/57/642557/3/check/nova-tox-functional/14e75d5/ : SUCCESS in 17m 41s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/3/check/nova-tox-functional-py35/7d8189b/ : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":3},{"id":"998c0f0c0a6aefc2cb6bf208aa95dfe5ccfa4a32","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:06:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/3/check/EMC_VxFlexOS_NOVA/c548539/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"4a5cd2c48387554e2a5366d60cced6cd59768a3b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:08:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43e53bcd3f04b9cac9d38ac370bee1b72900a5a0","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:54:06.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/642557/3/check/nova-out-of-tree-pvm/2d6d30e : FAILURE in 40m 42s\n- nova-in-tree-pvm http://184.172.12.213/57/642557/3/check/nova-in-tree-pvm/89f078b : FAILURE in 39m 30s","accounts_in_message":[],"_revision_number":3},{"id":"aa246297b76b3d7956f084e7f1d8c0dfffe4b455","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-21 11:03:30.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"ecdcadad4868a4381b43ab64390fd81f675e6866","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-21 12:52:02.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29436 : FAILURE in 1h 37m 55s","accounts_in_message":[],"_revision_number":3},{"id":"cbede7e7a4a1adb47391057cff7aaf0dfbac5969","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-21 14:35:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642557/3 : SUCCESS in 2h 28m 30s","accounts_in_message":[],"_revision_number":3},{"id":"cdfebc0d96b8cfd8f4885e2c7242e418dd6cac34","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-02 08:46:10.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e9022afcdf6e621279c7f5e7767787846f5301e5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 08:46:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8f91c6805322bbb0c683ff064075d949db9f919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 09:14:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/4/check/openstack-tox-cover/ad68686/cover/ : SUCCESS in 19m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/4/check/openstack-tox-lower-constraints/4f172d5/ : SUCCESS in 12m 58s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/4/check/openstack-tox-pep8/63eab4a/ : SUCCESS in 10m 01s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/4/check/openstack-tox-py27/1f32ef6/ : SUCCESS in 14m 07s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/4/check/openstack-tox-py35/3adc800/ : SUCCESS in 12m 52s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/4/check/openstack-tox-py36/b0e5ab3/ : SUCCESS in 15m 40s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/4/check/openstack-tox-py37/41c0bd5/ : SUCCESS in 13m 28s\n- openstack-tox-docs http://logs.openstack.org/57/642557/4/check/openstack-tox-docs/ea3221e/html/ : SUCCESS in 7m 20s\n- nova-tox-functional http://logs.openstack.org/57/642557/4/check/nova-tox-functional/79a6817/ : SUCCESS in 24m 50s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/4/check/nova-tox-functional-py35/16f4f9e/ : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":4},{"id":"c6e7fa0390aa7ae3540b1a0aaae4fdda20da410c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 09:15:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/4/check/EMC_VxFlexOS_NOVA/95a406f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"b9578dc1016d92049afd136717342d16398e03e5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 10:10:19.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642557/4 : FAILURE in 54s","accounts_in_message":[],"_revision_number":4},{"id":"0a4e7671c54deaa10a715013875294103f938ccc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-02 11:30:27.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"e651e240a798c706bf3bbb5ffb031643271406ad","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 13:20:27.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29860 : SUCCESS in 2h 10m 27s","accounts_in_message":[],"_revision_number":4},{"id":"243e9d72f3aecb9d3ccd30e9eb840faa3c5a5487","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-02 22:45:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI really want this series to land, but don\u0027t want to lose my +2 ability by fixing up the things that are bugging me. So I started a followup patch instead [1].\n\nIn it, I am not doing the things to \"reduce the diff\" because they don\u0027t make sense in the context of a fresh patch.\n\nAs for the unused mock args, rather than removing them with new\u003dmock.Mock(), I\u0027m adding assertions like a good boy.\n\nOnce I\u0027ve subsumed a given patch\u0027s updates that way, I\u0027ll be happy to approve said patch. In this case, the exercise so far has uncovered a real bug (see inline). I\u0027m not done going through this patch yet, so feel free to wait to fix.\n\nMeanwhile, if you update/rebase this series, please carry [1] along. (You should also feel free to take over that effort and/or squash it into the main series if you like.)\n\nThanks!\n\n[1] https://review.openstack.org/#/c/649471/","accounts_in_message":[],"_revision_number":4},{"id":"ff4611259cc4f2f5e15fa9be52d17cb5c27a6232","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-03 23:34:37.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/4/check-vote/ext-nova-zuul/e004ddc : FAILURE in 57m 37s","accounts_in_message":[],"_revision_number":4},{"id":"d53ddca98e210ba0da74b55013207994969a9ca3","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 08:46:40.000000000","message":"Patch Set 4:\n\nI do intend to get back to this one, but had been giving privsep priority. I\u0027m hoping to have the time to fixup this one tomorrow and will consume [1] as part of that.","accounts_in_message":[],"_revision_number":4},{"id":"74168e1f148696fb00356c0c60b84e839351695f","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-04 23:02:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e1a873af268ae368bd3f158e006c7e498e091f9","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-05 01:31:42.000000000","message":"Patch Set 1:\n\nOk, I\u0027ve done some tweaks. I\u0027ve given calls to file_open or write_file a pass given they\u0027re entirely removed later in the series.","accounts_in_message":[],"_revision_number":1},{"id":"7cabf20427d5830aa0ac37d566f3e520b3176d0c","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-05 09:53:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2136ba6ce18d4355384ab5cb9656fd7e7a450a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 10:23:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/5/check/openstack-tox-cover/f700b2a/cover/ : SUCCESS in 17m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/5/check/openstack-tox-lower-constraints/064f92f/ : SUCCESS in 13m 57s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/5/check/openstack-tox-pep8/fc7ee3a/ : SUCCESS in 9m 38s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/5/check/openstack-tox-py27/777b94e/ : SUCCESS in 12m 04s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/5/check/openstack-tox-py35/b93ca82/ : SUCCESS in 16m 04s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/5/check/openstack-tox-py36/3f13b02/ : SUCCESS in 14m 52s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/5/check/openstack-tox-py37/c38833b/ : SUCCESS in 10m 50s\n- openstack-tox-docs http://logs.openstack.org/57/642557/5/check/openstack-tox-docs/982399f/html/ : SUCCESS in 6m 23s\n- nova-tox-functional http://logs.openstack.org/57/642557/5/check/nova-tox-functional/6326164/ : SUCCESS in 18m 29s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/5/check/nova-tox-functional-py35/1f78999/ : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":5},{"id":"424b9400dc2415a0e1dbb51c8d9b3aee0d56a72e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 10:23:25.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/5/check/EMC_VxFlexOS_NOVA/d459a50/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"86cd953256d5bbcc3241c1ff7a71647cd777bfc8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-05 11:32:10.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30098 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":5},{"id":"326b37e97819797e5e24e7aac344b64d215926b2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-05 12:13:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642557/5 : SUCCESS in 2h 19m 23s","accounts_in_message":[],"_revision_number":5},{"id":"fda5976719d5b25f4ef5fdf4b94e51e109b99fd1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-05 13:06:35.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"451b43c80c7eb7d4ced1db9cc7a4ded15a38d67c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-06 00:04:10.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/5/check-vote/ext-nova-zuul/6f37a75 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":5},{"id":"138e8299f405a916bc7e2d150ce3fe0baa77d227","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-12 07:04:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"9a8caec8e91254cc2d3dd760d3ff7e86694b9b5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 07:59:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/6/check/openstack-tox-cover/737981b/cover/ : SUCCESS in 15m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/6/check/openstack-tox-lower-constraints/4214397/ : SUCCESS in 16m 53s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/6/check/openstack-tox-pep8/6d65528/ : SUCCESS in 11m 36s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/6/check/openstack-tox-py27/2496b8e/ : SUCCESS in 13m 48s\n- openstack-tox-py35 http://logs.openstack.org/57/642557/6/check/openstack-tox-py35/011dc8c/ : SUCCESS in 17m 05s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/6/check/openstack-tox-py36/709f52a/ : SUCCESS in 12m 31s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/6/check/openstack-tox-py37/7a52f4d/ : SUCCESS in 13m 59s\n- openstack-tox-docs http://logs.openstack.org/57/642557/6/check/openstack-tox-docs/0783c1c/html/ : SUCCESS in 6m 48s\n- nova-tox-functional http://logs.openstack.org/57/642557/6/check/nova-tox-functional/75f0fd4/ : SUCCESS in 20m 25s\n- nova-tox-functional-py35 http://logs.openstack.org/57/642557/6/check/nova-tox-functional-py35/7bb46cd/ : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":6},{"id":"15fccca9af5021b55b9da2241000dd1932360175","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 08:01:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/6/check/EMC_VxFlexOS_NOVA/7877d07/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"780ab85acb4c6ce0c7d3a48fe22624d645f83776","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-12 10:37:38.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642557/6 : FAILURE in 2m 07s","accounts_in_message":[],"_revision_number":6},{"id":"8a19e8aa4fd8d0c753537b2241b29f3f5a3c7827","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 13:40:53.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"27b6e8ec856b7699a395e819d2237a45f1331d4a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-12 16:57:11.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30438 : SUCCESS in 1h 34m 08s","accounts_in_message":[],"_revision_number":6},{"id":"163bfb2b55d7c5e23645a82c4cac32956ad21222","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-13 00:48:08.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"72e5620fbdde2d2e390f9112d2e4f4d27bc03f23","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-22 22:58:43.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"13f09a1a8b2af15111d3f18b8fe29c271062ec78","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-22 22:59:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0c81439dec8466218d7de379dd65f2633afdfbf3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-22 23:06:10.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30910 : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":7},{"id":"fd4481e05086e12dcf2a2c6b185b8e43a8792235","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:59:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/7/check/openstack-tox-cover/b64e22a/cover/ : SUCCESS in 15m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/7/check/openstack-tox-lower-constraints/5a0b568/ : SUCCESS in 12m 40s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/7/check/openstack-tox-pep8/ea3da61/ : SUCCESS in 9m 53s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/7/check/openstack-tox-py27/b9dd7a2/ : SUCCESS in 13m 41s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/7/check/openstack-tox-py36/f6bc367/ : SUCCESS in 12m 02s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/7/check/openstack-tox-py37/80638b2/ : SUCCESS in 13m 09s\n- openstack-tox-docs http://logs.openstack.org/57/642557/7/check/openstack-tox-docs/d753141/html/ : SUCCESS in 6m 18s\n- nova-tox-functional http://logs.openstack.org/57/642557/7/check/nova-tox-functional/847f487/ : SUCCESS in 20m 16s\n- nova-tox-functional-py36 http://logs.openstack.org/57/642557/7/check/nova-tox-functional-py36/36da8ae/ : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":7},{"id":"4728cdf859c4c9204bcda8dbb416cb302134daaa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-23 00:59:38.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/7/check/EMC_VxFlexOS_NOVA/c795646/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"203db5cc8993755e07ece989321e85cca33a9bb8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-23 01:17:45.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"a322372af68848c2c3f072da9dd63a75ca1930df","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-23 01:23:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642557/7 : SUCCESS in 2h 24m 00s","accounts_in_message":[],"_revision_number":7},{"id":"8404aa1f8ca9811e2d976e9f6e633b1588c9f6ee","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-23 02:06:36.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/642557/7/check/nova-out-of-tree-pvm/8e72e21 : FAILURE in 3h 07m 02s\n- nova-in-tree-pvm http://184.172.12.213/57/642557/7/check/nova-in-tree-pvm/42ceba3 : FAILURE in 2h 02m 32s","accounts_in_message":[],"_revision_number":7},{"id":"c846c53b46163cff43fffde4941207d1f0685126","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 12:00:23.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/7/check-vote/ext-nova-zuul/24cbd29 : FAILURE in 57m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c5b2f8cdb6761f4ae7446d538de7185b3a609a03","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 20:16:35.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\n/me finally gets back to this series...","accounts_in_message":[],"_revision_number":7},{"id":"f0e59b2a855bb7eaaad5703dd9e5b6945418c5af","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 20:27:48.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ef32124daa9a26d3c9bae1d22dd34dfd37a8d41b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:43:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/642557/8/check/openstack-tox-cover/69cdacb/cover/ : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/8/check/openstack-tox-lower-constraints/ac6ea1d/ : SUCCESS in 11m 46s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/8/check/openstack-tox-pep8/765050e/ : SUCCESS in 10m 15s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/8/check/openstack-tox-py27/9cb20dc/ : SUCCESS in 11m 48s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/8/check/openstack-tox-py36/53c9c10/ : SUCCESS in 16m 24s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/8/check/openstack-tox-py37/19fb19e/ : SUCCESS in 14m 22s\n- openstack-tox-docs http://logs.openstack.org/57/642557/8/check/openstack-tox-docs/9bbf634/html/ : SUCCESS in 7m 25s\n- nova-tox-functional http://logs.openstack.org/57/642557/8/check/nova-tox-functional/10b2d96/ : SUCCESS in 19m 11s\n- nova-tox-functional-py36 http://logs.openstack.org/57/642557/8/check/nova-tox-functional-py36/3f4a73e/ : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5301d919cc77b60e106f86b6b9dbd568ddd4dd10","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-13 22:43:42.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/642557/8/check/EMC_VxFlexOS_NOVA/bb0b80b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"6998d835e8a3eafc83b4876bcce8ed6549c00f19","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-14 00:05:12.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-642557-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"8f452e9282f153d4ed9ed70b2fe098446b44ed0e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-14 00:16:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ec7b8c0c26d135f4a067fd371989ba730d658bca","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-14 10:34:56.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642557/8 : SUCCESS in 2h 25m 00s","accounts_in_message":[],"_revision_number":8},{"id":"ae4d865b0e90c2db2a9f0dde353a8076ba5c2d65","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-05-14 13:53:11.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31820 : SUCCESS in 2h 22m 28s","accounts_in_message":[],"_revision_number":8},{"id":"68d8e9489fbab3f585be439cdeba768f9c72969a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 14:46:05.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/57/642557/8/check-vote/ext-nova-zuul/e7d8fac : FAILURE in 22m 23s","accounts_in_message":[],"_revision_number":8},{"id":"c88405202e25d921523b9be936b56e2218a9cd6a","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-05-16 19:53:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1115283fe35b0f14940b0cd542b63f0c6db982c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:54:04.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d8622af2de6c8ea7007f148f3566624b051688ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:38:34.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/642557/8/gate/openstack-tox-lower-constraints/d170307/ : SUCCESS in 14m 02s\n- openstack-tox-pep8 http://logs.openstack.org/57/642557/8/gate/openstack-tox-pep8/c337f87/ : SUCCESS in 10m 29s\n- openstack-tox-py27 http://logs.openstack.org/57/642557/8/gate/openstack-tox-py27/1cbdf66/ : SUCCESS in 12m 11s\n- openstack-tox-py36 http://logs.openstack.org/57/642557/8/gate/openstack-tox-py36/e7192f7/ : SUCCESS in 12m 25s\n- openstack-tox-py37 http://logs.openstack.org/57/642557/8/gate/openstack-tox-py37/9a8c449/ : SUCCESS in 14m 00s\n- openstack-tox-docs http://logs.openstack.org/57/642557/8/gate/openstack-tox-docs/52d3a10/html/ : SUCCESS in 6m 19s\n- nova-tox-functional http://logs.openstack.org/57/642557/8/gate/nova-tox-functional/16ebcb1/ : SUCCESS in 18m 09s\n- nova-tox-functional-py36 http://logs.openstack.org/57/642557/8/gate/nova-tox-functional-py36/157d447/ : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":8},{"id":"2209b8c505a459c9981c870768e253ec0c711495","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:38:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"83d9585f123b2d6824fdf111048b25e75792c125","revisions":{"d5601cdb9aeda7c0e4622ce4b2fa2a159841f994":{"kind":"REWORK","_number":1,"created":"2019-03-11 20:23:54.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/1"}}},"commit":{"parents":[{"commit":"28f2baca0294846a52763a5735c57b047ef6e483","subject":"Remove fake_libvirt_utils from the cache concurrency tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/28f2baca0294846a52763a5735c57b047ef6e483"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5601cdb9aeda7c0e4622ce4b2fa2a159841f994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5601cdb9aeda7c0e4622ce4b2fa2a159841f994"}]},"branch":"refs/heads/master"},"4efb163e40df70ca1df3cc2d6611f18088f11a22":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-03-18 09:22:28.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/2"}}},"commit":{"parents":[{"commit":"be8af28a4ff111f99da9b34ee36a90a8a25d5bef","subject":"Merge \"Trivial typo fix for REST API in policy enforcement docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/be8af28a4ff111f99da9b34ee36a90a8a25d5bef"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-18 06:02:50.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4efb163e40df70ca1df3cc2d6611f18088f11a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4efb163e40df70ca1df3cc2d6611f18088f11a22"}]},"branch":"refs/heads/master"},"cb6763a4e037c02d72700a0427e378401ab69de7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-21 09:37:58.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/3"}}},"commit":{"parents":[{"commit":"59f1f187e5dceb5841a711f265280346d70a972b","subject":"Merge \"docs: Misc cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59f1f187e5dceb5841a711f265280346d70a972b"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-21 09:37:31.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb6763a4e037c02d72700a0427e378401ab69de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb6763a4e037c02d72700a0427e378401ab69de7"}]},"branch":"refs/heads/master"},"c918b2ff5d8e910f179d8c7799dc2178cb6e4889":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-02 08:46:10.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/4"}}},"commit":{"parents":[{"commit":"2623acaba4573738faa61658af2d6006da6918a0","subject":"tests: Full stub out os_vif","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2623acaba4573738faa61658af2d6006da6918a0"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-02 03:32:58.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c918b2ff5d8e910f179d8c7799dc2178cb6e4889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c918b2ff5d8e910f179d8c7799dc2178cb6e4889"}]},"branch":"refs/heads/master"},"f9964a1718e25c97f3265903f096f2dc4e01db60":{"kind":"REWORK","_number":5,"created":"2019-04-05 09:53:32.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/5"}}},"commit":{"parents":[{"commit":"88560094450bb4935909793965866b542d9ef3fe","subject":"Add docs on what not to include in notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88560094450bb4935909793965866b542d9ef3fe"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-05 00:15:03.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9964a1718e25c97f3265903f096f2dc4e01db60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9964a1718e25c97f3265903f096f2dc4e01db60"}]},"branch":"refs/heads/master"},"0f0b2d26dc2c4ac47f35e83f16605cd8339866ad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-04-12 07:04:36.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/6"}}},"commit":{"parents":[{"commit":"03322bb517925a9f5a04ebdb41c3fd31e7962440","subject":"Merge \"trivial: Remove dead nova.db functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03322bb517925a9f5a04ebdb41c3fd31e7962440"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-12 06:58:51.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f0b2d26dc2c4ac47f35e83f16605cd8339866ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f0b2d26dc2c4ac47f35e83f16605cd8339866ad"}]},"branch":"refs/heads/master"},"1302a414108685125c1b1c28531833eaa967cc15":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-04-22 22:58:43.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/57/642557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/7"}}},"commit":{"parents":[{"commit":"fc3890667e4971e3f0f35ac921c2a6c25f72adec","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc3890667e4971e3f0f35ac921c2a6c25f72adec"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-22 22:14:31.000000000","tz":0},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1302a414108685125c1b1c28531833eaa967cc15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1302a414108685125c1b1c28531833eaa967cc15"}]},"branch":"refs/heads/master"},"83d9585f123b2d6824fdf111048b25e75792c125":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-05-13 20:27:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/57/642557/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/642557/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/642557/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/642557/8"}}},"commit":{"parents":[{"commit":"780f737547ee0b8f39f0bc1e4d6c4de6f8d1dd54","subject":"Merge \"api-ref: \u0027os-hypervisors\u0027 doesn\u0027t reflect overcommit ratio\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/780f737547ee0b8f39f0bc1e4d6c4de6f8d1dd54"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:56:09.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-13 20:25:38.000000000","tz":-300},"subject":"Remove fake_libvirt_utils from connection tests.","message":"Remove fake_libvirt_utils from connection tests.\n\nThe fake implementation of libvirt.utils is confusing to debug\nwhen things go wrong, direct mocks are clearer and keep everything\nin one place.\n\nChange-Id: Ief2c959d7c339e11b19e521de6e7d1be2f8722f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83d9585f123b2d6824fdf111048b25e75792c125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83d9585f123b2d6824fdf111048b25e75792c125"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
