)]}'
{"id":"openstack%2Fnova~642558","triplet_id":"openstack%2Fnova~master~Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd","project":"openstack/nova","branch":"master","topic":"fake-cleanup","hashtags":[],"change_id":"Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd","subject":"Remove fake_libvirt_utils from snapshot tests.","status":"MERGED","created":"2019-03-11 20:23:54.000000000","updated":"2019-05-31 12:05:43.000000000","submitted":"2019-05-31 12:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"642558-1559304343466-db54cbaf","meta_rev_id":"cba6ca3257695de0f64073981d6398151f95c596","_number":642558,"virtual_id_number":642558,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-14 14:49:04.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 00:06:24.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 10:31:59.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-05-13 22:45:23.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 00:16:43.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-14 12:52:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-05-31 12:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-13 20:27:48.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-31 10:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-05-31 10:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-11 20:24:06.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-23 02:07:41.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 20:20:12.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-13 22:45:23.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 00:06:24.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 00:16:43.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 10:31:59.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-14 12:52:16.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-05-14 14:49:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-31 10:59:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-31 12:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a01ca9c69d5135462e5d237a83cec418232267a8","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-11 20:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdacda5bc0c4f9042e7e1dfc0d93c9fa0d6a745d","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-03-11 20:24:06.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/642558/1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cc0e6e810a808b5975d61018ce4fe005795972c8","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-11 20:26:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"712edba496f5dcf520abde40bed0fd2516733203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 20:50:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/1/check/openstack-tox-cover/3bdca8b/cover/ : SUCCESS in 21m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/1/check/openstack-tox-lower-constraints/7fa7154/ : SUCCESS in 12m 20s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/1/check/openstack-tox-pep8/98c785e/ : SUCCESS in 11m 11s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/1/check/openstack-tox-py27/655504d/ : SUCCESS in 14m 43s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/1/check/openstack-tox-py35/3ad76be/ : SUCCESS in 13m 26s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/1/check/openstack-tox-py36/d2a434c/ : SUCCESS in 11m 43s\n- openstack-tox-docs http://logs.openstack.org/58/642558/1/check/openstack-tox-docs/3e9f79d/html/ : SUCCESS in 8m 13s\n- nova-tox-functional http://logs.openstack.org/58/642558/1/check/nova-tox-functional/be9d093/ : SUCCESS in 20m 16s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/1/check/nova-tox-functional-py35/d9cc9d4/ : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7fe5b8392ba6a910006c90be59f0749f52ca25d3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-11 20:59:02.000000000","message":"Patch Set 1:\n\nNice. Once again, I would prefer avoiding extra test method args where they\u0027re not used. And once again, let me know if you think that\u0027s horrible and wrong.","accounts_in_message":[],"_revision_number":1},{"id":"9f8c04ffe49a95b215c027c1fa152df8db93bf50","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-11 22:06:01.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29073 : SUCCESS in 1h 38m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c8758c6b3c76e0724223e25f3dbc6c396440e978","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-11 22:39:19.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642558/1 : FAILURE in 2h 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f40f49b82ea272cb67ff996cacd13349a1ac2d2d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-11 22:44:48.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/1/check-vote/ext-nova-zuul/71825d3 : FAILURE in 54m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4627cb0441afb8d02d1185314265f2abd466f7f9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-12 01:25:31.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"f6a48e672015ee6fc4af67c5fe0fa0fd0372b63b","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-18 09:22:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3137c43ba8f96befbd795552f7d8902939da2a1a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 09:23:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46345fae2e12eb75ec22264082097607e51fe6cf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-18 09:36:08.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29298 : FAILURE in 9m 06s","accounts_in_message":[],"_revision_number":2},{"id":"24c8f254bc736a87c6c5b7bb42f131fdad1feb17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 09:44:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/2/check/openstack-tox-cover/91d9e36/cover/ : SUCCESS in 18m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/2/check/openstack-tox-lower-constraints/ac2de72/ : SUCCESS in 12m 49s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/2/check/openstack-tox-pep8/3918617/ : SUCCESS in 8m 44s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/2/check/openstack-tox-py27/e9916f0/ : SUCCESS in 15m 20s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/2/check/openstack-tox-py35/1bc9228/ : SUCCESS in 12m 31s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/2/check/openstack-tox-py36/964161e/ : SUCCESS in 12m 29s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/2/check/openstack-tox-py37/18efa18/ : SUCCESS in 10m 53s\n- openstack-tox-docs http://logs.openstack.org/58/642558/2/check/openstack-tox-docs/efdb1ad/html/ : SUCCESS in 8m 22s\n- nova-tox-functional http://logs.openstack.org/58/642558/2/check/nova-tox-functional/1362c34/ : SUCCESS in 16m 28s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/2/check/nova-tox-functional-py35/764637b/ : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f9bf66b32fe1237face088d1196724873e38fc8c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-03-18 11:49:41.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/2/check-vote/ext-nova-zuul/83d4c4b : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":2},{"id":"15e72a9eb7db324e0d782c8f70c3f8626ce6ee91","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-18 13:12:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642558/2 : SUCCESS in 2h 42m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7bc7124d1be8cdcdca331a428a61d617816b21aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-18 15:41:56.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"2b4423e3a39a34ed86bdc50990a53ca1ac08b3bc","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-03-21 09:37:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"53cd2aaa43cb60dc66d45ac4fd4d62437e686582","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:08:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e9c458ca8c01e4a58f3b80ded5253b27f8c37b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 10:10:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/3/check/openstack-tox-cover/0ba2226/cover/ : SUCCESS in 18m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/3/check/openstack-tox-lower-constraints/4f638dc/ : SUCCESS in 14m 17s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/3/check/openstack-tox-pep8/dfa1d41/ : SUCCESS in 10m 14s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/3/check/openstack-tox-py27/d8349e7/ : SUCCESS in 13m 41s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/3/check/openstack-tox-py35/a3f6580/ : SUCCESS in 12m 53s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/3/check/openstack-tox-py36/a79f925/ : SUCCESS in 14m 52s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/3/check/openstack-tox-py37/6b683a6/ : SUCCESS in 12m 48s\n- openstack-tox-docs http://logs.openstack.org/58/642558/3/check/openstack-tox-docs/dc8f9f2/html/ : SUCCESS in 6m 50s\n- nova-tox-functional http://logs.openstack.org/58/642558/3/check/nova-tox-functional/c7d78f7/ : SUCCESS in 18m 50s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/3/check/nova-tox-functional-py35/3c74a45/ : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":3},{"id":"7c349a67b9892b88a934312a5f06f02c93a58cc6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:10:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/58/642558/3/check/EMC_VxFlexOS_NOVA/9bb7004/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"527046717485eb65a187b1e9605e2f557d88750b","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-21 10:54:54.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/58/642558/3/check/nova-out-of-tree-pvm/73c71b2 : FAILURE in 40m 11s\n- nova-in-tree-pvm http://184.172.12.213/58/642558/3/check/nova-in-tree-pvm/04e3b24 : FAILURE in 39m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5beaf538f09c1d21f8c5a3306efaef9838aff82d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-21 12:10:40.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"b644cc2a945eb69eeca99865cc087ce3ea95e36c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-21 12:20:44.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642558/3 : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":3},{"id":"0a1f4375ab4dd2d0aa1df5fb46521db4812b1a33","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-21 13:38:04.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29437 : SUCCESS in 1h 59m 56s","accounts_in_message":[],"_revision_number":3},{"id":"928bcff390235615c3d186cd6ccc4953c8202cdf","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-02 08:46:10.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"76e5cf1147ea17b84f63cc664896fa5b60d6c041","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 08:47:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4553ef0133c8341df0674c4b4a1c7c7dc2e6fb9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 09:21:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/4/check/openstack-tox-cover/13fed90/cover/ : SUCCESS in 15m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/4/check/openstack-tox-lower-constraints/6871603/ : SUCCESS in 13m 30s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/4/check/openstack-tox-pep8/fe306b0/ : SUCCESS in 9m 49s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/4/check/openstack-tox-py27/4511e92/ : SUCCESS in 13m 46s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/4/check/openstack-tox-py35/99797f9/ : SUCCESS in 14m 07s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/4/check/openstack-tox-py36/9155ec0/ : SUCCESS in 23m 00s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/4/check/openstack-tox-py37/b4819a2/ : SUCCESS in 23m 54s\n- openstack-tox-docs http://logs.openstack.org/58/642558/4/check/openstack-tox-docs/896bf42/html/ : SUCCESS in 7m 59s\n- nova-tox-functional http://logs.openstack.org/58/642558/4/check/nova-tox-functional/7ddcf15/ : SUCCESS in 20m 17s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/4/check/nova-tox-functional-py35/169132a/ : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":4},{"id":"942f97256193a62b619e1743410990a3e1503cb4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 09:21:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/58/642558/4/check/EMC_VxFlexOS_NOVA/0a6f517/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"b5427615346b830e965853fd46ee2f492cb13137","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 10:10:25.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642558/4 : FAILURE in 50s","accounts_in_message":[],"_revision_number":4},{"id":"bcca671f622c698fc0ee7926306759e8aa2af836","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 12:56:16.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29861 : SUCCESS in 1h 30m 22s","accounts_in_message":[],"_revision_number":4},{"id":"1326d7f38d9859f231aa57d33d5c2fa443d419e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-04 00:29:45.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/4/check-vote/ext-nova-zuul/8d4ce25 : FAILURE in 1h 03m 45s","accounts_in_message":[],"_revision_number":4},{"id":"dc7357f157ea83179960eef1c261bc6e29a8a253","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-05 09:53:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e546c8d54441bb2daf1edf9f00e337356d06fa65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 10:21:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/5/check/openstack-tox-cover/300b55f/cover/ : SUCCESS in 15m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/5/check/openstack-tox-lower-constraints/3eb296a/ : SUCCESS in 16m 11s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/5/check/openstack-tox-pep8/f9de173/ : SUCCESS in 9m 14s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/5/check/openstack-tox-py27/37c7774/ : SUCCESS in 23m 28s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/5/check/openstack-tox-py35/cb758ee/ : SUCCESS in 12m 40s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/5/check/openstack-tox-py36/43c6cdb/ : SUCCESS in 23m 05s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/5/check/openstack-tox-py37/7887d8f/ : SUCCESS in 11m 14s\n- openstack-tox-docs http://logs.openstack.org/58/642558/5/check/openstack-tox-docs/3fdd401/html/ : SUCCESS in 6m 08s\n- nova-tox-functional http://logs.openstack.org/58/642558/5/check/nova-tox-functional/f3d6e39/ : SUCCESS in 16m 44s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/5/check/nova-tox-functional-py35/bf9124b/ : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":5},{"id":"fb3d98dfb2905661d8ec875c7d2e3648a0a21d91","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-05 10:22:22.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/58/642558/5/check/EMC_VxFlexOS_NOVA/6ff0703/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"49941c968b1f4fa70d25ee04d2243075d07b2bf9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-05 11:30:48.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30099 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7c7f819d11310732f399d40175f6d0a2e562a0e0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-05 12:16:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642558/5 : SUCCESS in 2h 21m 43s","accounts_in_message":[],"_revision_number":5},{"id":"d97c07178993a644a9682fc1136c39243e0a42ec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-05 13:06:21.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"3efe3ee3445a93b0fcb139bed5cdd89e15218eb0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-06 00:49:19.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/5/check-vote/ext-nova-zuul/87ddc2e : FAILURE in 1h 30m 47s","accounts_in_message":[],"_revision_number":5},{"id":"60250008dca96d8c2ccfd159c05fee8177e2d1fe","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-12 07:04:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3890cc1ae7db0f4b74d531daed2492a70f6e40f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 08:07:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/6/check/openstack-tox-cover/dcbd8a8/cover/ : SUCCESS in 18m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/6/check/openstack-tox-lower-constraints/ba6cafd/ : SUCCESS in 23m 31s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/6/check/openstack-tox-pep8/3e1492e/ : SUCCESS in 10m 38s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/6/check/openstack-tox-py27/291c7e4/ : SUCCESS in 13m 46s\n- openstack-tox-py35 http://logs.openstack.org/58/642558/6/check/openstack-tox-py35/cf84688/ : SUCCESS in 14m 24s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/6/check/openstack-tox-py36/a167e46/ : SUCCESS in 13m 01s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/6/check/openstack-tox-py37/c2f1171/ : SUCCESS in 12m 43s\n- openstack-tox-docs http://logs.openstack.org/58/642558/6/check/openstack-tox-docs/31a9c18/html/ : SUCCESS in 8m 00s\n- nova-tox-functional http://logs.openstack.org/58/642558/6/check/nova-tox-functional/568c966/ : SUCCESS in 20m 38s\n- nova-tox-functional-py35 http://logs.openstack.org/58/642558/6/check/nova-tox-functional-py35/b4fd7f2/ : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":6},{"id":"438cac92012c4c50204155ef7e20ec7bca1132f5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 08:07:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/58/642558/6/check/EMC_VxFlexOS_NOVA/c9a0f5f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"52fe11763714280e37328434cf9b275db34e27a0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-12 10:55:00.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642558/6 : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":6},{"id":"91c627c4244aef4a8c3a2be722f2c0cabcc24f33","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-12 13:41:14.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"4a061650a892c565bda52cbf8706ecf316dd0ac6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-12 17:37:42.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30439 : SUCCESS in 2h 09m 16s","accounts_in_message":[],"_revision_number":6},{"id":"550eb2aab516f79905da6aad84a562971407f3d1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-13 04:23:52.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"7db41ebb6f2d0fb7cb26fa209099b1c9d171a728","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2019-04-22 22:58:43.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7d1b4d51d29ee6e09258f0a54a8497493729e49d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-22 22:59:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"52ae3311fd4f88b0448ca14fea13ad070ad30027","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-22 23:05:51.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30911 : FAILURE in 3m 45s","accounts_in_message":[],"_revision_number":7},{"id":"a08791fbe1308c8c4111102de58eadd156595de2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:01:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/7/check/openstack-tox-cover/7ae2292/cover/ : SUCCESS in 17m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/7/check/openstack-tox-lower-constraints/93c9703/ : FAILURE in 14m 58s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/7/check/openstack-tox-pep8/221827a/ : SUCCESS in 10m 37s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/7/check/openstack-tox-py27/208d95c/ : SUCCESS in 24m 09s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/7/check/openstack-tox-py36/0cd34a5/ : SUCCESS in 13m 51s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/7/check/openstack-tox-py37/a81c54a/ : SUCCESS in 23m 19s\n- openstack-tox-docs http://logs.openstack.org/58/642558/7/check/openstack-tox-docs/e441db0/html/ : SUCCESS in 7m 07s\n- nova-tox-functional http://logs.openstack.org/58/642558/7/check/nova-tox-functional/5c03f8d/ : SUCCESS in 19m 56s\n- nova-tox-functional-py36 http://logs.openstack.org/58/642558/7/check/nova-tox-functional-py36/3f8c55d/ : SUCCESS in 21m 57s","accounts_in_message":[],"_revision_number":7},{"id":"fad663b3192ea35c09c91424a0486549e9ee96aa","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-23 01:12:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/642558/7 : SUCCESS in 2h 12m 38s","accounts_in_message":[],"_revision_number":7},{"id":"f94a639ae7bde8085b006c6d0444708916c429b2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-23 02:07:41.000000000","message":"Patch Set 7:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/58/642558/7/check/nova-out-of-tree-pvm/bcb0b22 : FAILURE in 3h 07m 33s\n- nova-in-tree-pvm http://184.172.12.213/58/642558/7/check/nova-in-tree-pvm/01f954e : FAILURE in 2h 28m 18s","accounts_in_message":[],"_revision_number":7},{"id":"d2f96c1cc9dfa3f40d189c92878fa430180698ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-23 04:40:19.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/7/check-vote/ext-nova-zuul/04ce952 : FAILURE in 1h 02m 52s","accounts_in_message":[],"_revision_number":7},{"id":"5a6791814dc20b94ec230670cee9f717cefb6817","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 20:20:12.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"d937a084fb4f82c3856accefb9854629d935f3bb","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-13 20:27:48.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"1aa09735d4db07a3c63559cf0de4ce69e1ff25be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 22:44:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/58/642558/8/check/openstack-tox-cover/27a80e3/cover/ : SUCCESS in 16m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/8/check/openstack-tox-lower-constraints/8d5b2fe/ : SUCCESS in 14m 47s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/8/check/openstack-tox-pep8/7800a31/ : SUCCESS in 10m 27s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/8/check/openstack-tox-py27/4a62eff/ : SUCCESS in 11m 09s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/8/check/openstack-tox-py36/54eb65c/ : SUCCESS in 14m 43s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/8/check/openstack-tox-py37/08ef8cd/ : SUCCESS in 15m 09s\n- openstack-tox-docs http://logs.openstack.org/58/642558/8/check/openstack-tox-docs/6157684/html/ : SUCCESS in 5m 41s\n- nova-tox-functional http://logs.openstack.org/58/642558/8/check/nova-tox-functional/5265104/ : SUCCESS in 22m 16s\n- nova-tox-functional-py36 http://logs.openstack.org/58/642558/8/check/nova-tox-functional-py36/f10582e/ : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":8},{"id":"660b4f89e16cdad04d4ae3221733b0d23b13043e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-05-13 22:45:23.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/58/642558/8/check/EMC_VxFlexOS_NOVA/ecee9df/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"9d060e86b6d3d4c667aaf07cecd80c31b15a8df0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-14 00:06:24.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-58-642558-8 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":8},{"id":"20532bde686832e8c68ce3b7b7017d2bdbea1042","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-05-14 00:16:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8c8820ecbe06f6fb08b7f484bec3e03ad2fb92a1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-14 10:31:59.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/642558/8 : FAILURE in 2h 21m 24s","accounts_in_message":[],"_revision_number":8},{"id":"36ea19ae0ebf38f549b9ff30e73b0183985f66fc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-05-14 12:52:16.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31821 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":8},{"id":"eab2fe3e94d3b597071771239398821c5b706c60","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-05-14 14:49:04.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/58/642558/8/check-vote/ext-nova-zuul/35ea4ce : FAILURE in 24m 38s","accounts_in_message":[],"_revision_number":8},{"id":"0c5fab211cc607c9241bdfa4aba3b0250ba61d4b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-31 10:59:45.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4a32e4d76ec92cba68bb76e1a091128881f17391","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:01:11.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8d859f193d89fdcba80509fe0541fb9731c815bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:05:42.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/58/642558/8/gate/openstack-tox-lower-constraints/002b285/ : SUCCESS in 17m 15s\n- openstack-tox-pep8 http://logs.openstack.org/58/642558/8/gate/openstack-tox-pep8/11e9bd3/ : SUCCESS in 9m 59s\n- openstack-tox-py27 http://logs.openstack.org/58/642558/8/gate/openstack-tox-py27/a15e38e/ : SUCCESS in 17m 03s\n- openstack-tox-py36 http://logs.openstack.org/58/642558/8/gate/openstack-tox-py36/3f579fe/ : SUCCESS in 16m 22s\n- openstack-tox-py37 http://logs.openstack.org/58/642558/8/gate/openstack-tox-py37/3394890/ : SUCCESS in 12m 29s\n- openstack-tox-docs http://logs.openstack.org/58/642558/8/gate/openstack-tox-docs/b16bca0/html/ : SUCCESS in 7m 34s\n- nova-tox-functional http://logs.openstack.org/58/642558/8/gate/nova-tox-functional/8a0484e/ : SUCCESS in 18m 09s\n- nova-tox-functional-py36 http://logs.openstack.org/58/642558/8/gate/nova-tox-functional-py36/0904ba9/ : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":8},{"id":"e364008503aad4c354716c869493e006f19da124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:05:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d43cab00a6c18dd3778f0a4e83ec5e753ab498ab","revisions":{"8d163c5f990ba43dba64a1e8af09eefcce8ae519":{"kind":"REWORK","_number":1,"created":"2019-03-11 20:23:54.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/1"}}},"commit":{"parents":[{"commit":"d5601cdb9aeda7c0e4622ce4b2fa2a159841f994","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5601cdb9aeda7c0e4622ce4b2fa2a159841f994"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 09:55:54.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d163c5f990ba43dba64a1e8af09eefcce8ae519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d163c5f990ba43dba64a1e8af09eefcce8ae519"}]},"branch":"refs/heads/master"},"4bd732d30e47d583a07b36266f8e610ffe555956":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-03-18 09:22:28.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/2"}}},"commit":{"parents":[{"commit":"4efb163e40df70ca1df3cc2d6611f18088f11a22","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4efb163e40df70ca1df3cc2d6611f18088f11a22"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-18 06:02:50.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bd732d30e47d583a07b36266f8e610ffe555956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bd732d30e47d583a07b36266f8e610ffe555956"}]},"branch":"refs/heads/master"},"30fc92b7db125f4b5378454dce74520a97048a5e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-03-21 09:37:58.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/3"}}},"commit":{"parents":[{"commit":"cb6763a4e037c02d72700a0427e378401ab69de7","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb6763a4e037c02d72700a0427e378401ab69de7"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-21 09:37:31.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30fc92b7db125f4b5378454dce74520a97048a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30fc92b7db125f4b5378454dce74520a97048a5e"}]},"branch":"refs/heads/master"},"507fc1de245e02826d72389a76958e582e79ed4a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-02 08:46:10.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/4"}}},"commit":{"parents":[{"commit":"c918b2ff5d8e910f179d8c7799dc2178cb6e4889","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c918b2ff5d8e910f179d8c7799dc2178cb6e4889"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-02 03:32:58.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/507fc1de245e02826d72389a76958e582e79ed4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/507fc1de245e02826d72389a76958e582e79ed4a"}]},"branch":"refs/heads/master"},"1ce0960f4868f9dca643112eb03cdafa42018085":{"kind":"REWORK","_number":5,"created":"2019-04-05 09:53:32.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/5"}}},"commit":{"parents":[{"commit":"f9964a1718e25c97f3265903f096f2dc4e01db60","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9964a1718e25c97f3265903f096f2dc4e01db60"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-05 09:20:38.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ce0960f4868f9dca643112eb03cdafa42018085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ce0960f4868f9dca643112eb03cdafa42018085"}]},"branch":"refs/heads/master"},"5d8faf7b46082b88260564a686f8ecb1d54f7fd4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-04-12 07:04:36.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/6"}}},"commit":{"parents":[{"commit":"0f0b2d26dc2c4ac47f35e83f16605cd8339866ad","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f0b2d26dc2c4ac47f35e83f16605cd8339866ad"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-12 06:58:51.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d8faf7b46082b88260564a686f8ecb1d54f7fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5d8faf7b46082b88260564a686f8ecb1d54f7fd4"}]},"branch":"refs/heads/master"},"c90f55bf2fe21b94336990db5595163ab6c4dd48":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-04-22 22:58:43.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/58/642558/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/7"}}},"commit":{"parents":[{"commit":"1302a414108685125c1b1c28531833eaa967cc15","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1302a414108685125c1b1c28531833eaa967cc15"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-04-22 22:14:31.000000000","tz":0},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c90f55bf2fe21b94336990db5595163ab6c4dd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c90f55bf2fe21b94336990db5595163ab6c4dd48"}]},"branch":"refs/heads/master"},"d43cab00a6c18dd3778f0a4e83ec5e753ab498ab":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-05-13 20:27:48.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/58/642558/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/642558/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/642558/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/642558/8"}}},"commit":{"parents":[{"commit":"83d9585f123b2d6824fdf111048b25e75792c125","subject":"Remove fake_libvirt_utils from connection tests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83d9585f123b2d6824fdf111048b25e75792c125"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2019-03-11 06:57:37.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-13 20:25:38.000000000","tz":-300},"subject":"Remove fake_libvirt_utils from snapshot tests.","message":"Remove fake_libvirt_utils from snapshot tests.\n\nThis is probably the most complicated transition for this file.\nHowever, it makes it much clearer what each test expects in order\nto work correctly.\n\nChange-Id: Iddd5f1f31a7bdcfee94b5113af0b2b25fe677edd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d43cab00a6c18dd3778f0a4e83ec5e753ab498ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d43cab00a6c18dd3778f0a4e83ec5e753ab498ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
